)]}'
{"id":"openstack%2Fcinder~923246","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I65857288b797cde573e7443ac6e7e6f57fedde01","project":"openstack/cinder","branch":"stable/2023.2","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-04 21:36:46.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-04 21:36:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65857288b797cde573e7443ac6e7e6f57fedde01","subject":"CVE-2024-32498: Check for external qcow2 data file","status":"MERGED","created":"2024-07-02 14:04:08.000000000","updated":"2024-07-05 10:06:01.000000000","submitted":"2024-07-04 21:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923246-bug/2059809","meta_rev_id":"133aa4ede0b97e9681504f87e351e60b7cb8c684","_number":923246,"virtual_id_number":923246,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 21:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2024-07-03 20:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-03 09:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2024-07-03 20:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:04:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 14:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 14:52:55.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-02 16:14:44.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-02 16:15:02.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-02 16:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 09:11:05.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-07-03 09:43:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-07-03 15:25:21.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-03 16:41:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-03 20:46:46.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-04 05:34:42.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-07-05 07:30:22.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"7b5315862f9495963bcf75246d13cd3465d4a147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-02 14:04:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3153fac2d383461b093e949b81202f6366797df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-02 14:04:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"3a78ca7e73c16fcd01c1a374924c2f3499a23b47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:41:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90e67b391b148d9bcd0c800880d53ce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1065e804dd1469a9728fb0484edbcbd : FAILURE in 3m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ada241f99de2464db71403d56d56ce8c : FAILURE in 3m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b52d7ecca3e4fe5b0b5f9eab0ca2c61 : FAILURE in 3m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e48d07be11b3489481c50cf4c85c3f76 : FAILURE in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bad27e9f89a87ccad386aec574955860ded4233","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-02 14:52:55.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/39c1243fe5cc49f0a10ad5c925374826\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a6b3d45b0c0f42c1bd24a7f6302f2e64 : POST_FAILURE in 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3733294b7bbc7fe5f1f126931dc3ce294a383102","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-02 16:14:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923246 : FAILURE in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0ed5d0b7385eff667cedcdf0df888cb9c9f7bf92","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-02 16:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923246 : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ddead845b86aae4d5ed33bbfb7a3344166c46519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 16:46:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6b5eeaea35d4240a660df984062bb88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81adf76ec7bc49a28d2a9bb9faf8cb78 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f5ad3f003441ae8506fa53a00e8436 : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f5c71254f294e2794d69bb6ec2b74d4 : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26fd8da9acfc475fb11f7904e4df292d : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff7100c5c0db4112ba94dc65a4df8454 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eb939713e84b4085a463875a11b1aa : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4363179b0cc54d9e99a004ecf362c83e : SUCCESS in 1h 01m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8729997bdee44ee3915ae80d27a0785b : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7fff3eb37784a058ee0c30afdc52898 : SUCCESS in 1h 07m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56e1b8851bcc41f8b297b7dd8886d295 : SUCCESS in 17m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9952d72d540646b79b44b261711e53ed : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cdec2d99193a4250ad9cf1ff61e41be7 : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b336e869541431d8b410f666168155c : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f38b041340b47978d4c758d8e3562e5 : SUCCESS in 6m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66b87b4410bc4e95914bb788dffa30c5 : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9572a80c2e234a498f9956f497ff0e18 : SUCCESS in 40m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5e2da4b21ad4d679986bb07d5fa694d : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb1d942b80f74411ba232477293a15f6 : SUCCESS in 1h 56m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e384a7e9bcd4f49aff80389d2e937dd : FAILURE in 2h 09m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d7646d15c54946dfb544c05dba11997b : FAILURE in 2h 09m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7870f6b95a94b9ba1f861d6dec2b65f : FAILURE in 2h 32m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5613d784a1ef4b1bbf920f8f5d8ce6f3 : SUCCESS in 1h 24m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0104f945cbd248cea25e33106e354ec8 : FAILURE in 2h 06m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d3caed16cec45e5a367e8dd3ab66155 : FAILURE in 51m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/74cded7caf6843808d07d339209f133c : SUCCESS in 1h 20m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/142a2a7c7c35449ab0165c9560dc6799 : SUCCESS in 1h 40m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6824bccc7ba4139bde3a794515694ae : FAILURE in 1h 47m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6e45be7d15c43d8abb4db221c9d7c15 : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d6669f830b5762b1f55a09b8bda7d85b36749761","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-03 09:11:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a4211bd079573fccffb26e971d28f12b0b42853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-03 09:15:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d32f4818e93a0a2db8687df16bb055d5c09faddb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-03 09:16:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"be2c017be3feb52a3644c27406ea0428373eef5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:42:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56985b522dc941cb86fd0b9dc2fcf3f0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05639bb1836c46cd87a4127b8336d42b : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1e9dcb9b0744b90beb5fde5197dfdd9 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b4ed46ddd974c5e8727c5368e193ac2 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3db6d2646604f09b5f061b9c1679e4e : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a9158239398048d417361534356661dfa1ed4e6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-03 09:43:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c163119ccf331952e7860790d876f69110a81d55","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-03 09:48:35.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a1b09aadb77c4048906c2bbf4b5dab1e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3617017f09a444f29c8462654b7bbf36 : POST_FAILURE in 24m 21s","accounts_in_message":[],"_revision_number":2},{"id":"20e7ed18589c3e77af0a915a3f743d46a2b3954e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-03 09:56:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923246 : FAILURE in 40m 03s","accounts_in_message":[],"_revision_number":2},{"id":"5a86fbb93c40edbf30161522267693d5cf90981d","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-03 10:00:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923246 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d529fd15efd8e0f708734f0661bd863a3da85ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 12:10:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48fc62c88f342508229a461b4d53879\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f1cc41bf8f46218459d3d02f3758ee : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cff06d3a46824e04a3df9498128d6aea : FAILURE in 13m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f3631b996ef4528beeebddf246ba5de : SUCCESS in 11m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de26bf8bb9c34231bdbc116f414421a9 : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c33ee33e33924ae09a4a5bc697f1ad10 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd70364fd62a4c2191935da2d53669a2 : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/d1bddd915ae049709318a6d9c1aa5adf : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a80b2a9778964a6ea5509a16f9e40c2e : SUCCESS in 1h 19m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df3bcc4111d54b68bc15e88192d2b9df : SUCCESS in 1h 00m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78c28758e62a493fb44f2465fb95506f : SUCCESS in 17m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6034a397bb45416896e8b4866677b891 : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22d0192f5ec14533af66cc28c16f2b56 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11acdeb5ba8c49c5a39ce6fed2b5780c : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9652a725991d4724af74c6880e7ec0dc : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b0fc51885f44705b05c03d6acdffd31 : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/047c56e5447f4c2096a8ad5240827c17 : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1a806be0d144e2793a5416ca2ab8e55 : SUCCESS in 7m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a3228755d5c49518a8d4feba1ef8a3a : SUCCESS in 1h 56m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/393129d7694a4d929a639f77cc54bc6b : SUCCESS in 2h 00m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5326057eeb1a4ada9f1d1ba7d8b346e2 : SUCCESS in 1h 43m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ab6624cf490f4bfbac3561e708eddf5c : SUCCESS in 2h 08m 17s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0058003b37e0467abd8c45ce100fa6a5 : SUCCESS in 1h 07m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/085515cc79fa47a2ab54467c4f9743c5 : SUCCESS in 1h 40m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bc79d3bb9da847b29032eec068fc633f : SUCCESS in 32m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25cc40c925a44c07a4c6298a98b75973 : SUCCESS in 1h 25m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cc0600dabec84326b259e10521777bcb : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0168ebb3f9ee4c599cf5d2594750b015 : SUCCESS in 1h 40m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a681ce5036b24b448d52bd3fe394e0af : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":2},{"id":"eaca95ca6ab3f6cf21cd2174fc6af1966b3b6d65","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-03 12:39:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94b23145d649409ef3be05c936eed49e1148663b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-03 13:02:48.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c66854a84b1c4ff2a19e08f76060b6a9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/44e523790f774f6f8d6189ccf8810843 : POST_FAILURE in 20m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a8bfc0126bdd622aa3fe22cfbcbf486eb45c5656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 13:03:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e9893a2bd0a41789c10e736df960209\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bae806d230314686afa7e8e39d25572a : FAILURE in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1de69c63e8c048d59444d641c6c51769 : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de2e2383fa55402f927d1699099bd359 : FAILURE in 22m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a93f821b66cc4da0941bc68eec75426e : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f1bc80d1aef8bf5ae5e37d9cd04270a485bd56c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 15:19:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5cf39b3969475ea406959088052a8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06da5da9d3c14675a5a2a7f88b38c805 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0ca9ab0e654403b93898bc4a620b67f : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d3e0d68869c4e2499ba0cc460680979 : SUCCESS in 11m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e336f60c5b78450aa889b923fbd1beeb : SUCCESS in 10m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8de0daa852f245519e65430af97d2af0 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ce237493e34a448abd54590d0ce0d1 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f88ab7d21fd344d2a02e583415621a3f : SUCCESS in 56m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae70e44e2d6f46cba526a60229e672a3 : SUCCESS in 1h 25m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf89ce31106d4d3789bce42519c3f260 : SUCCESS in 57m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33371b2c2dd64ba4b14f18bafa5c23e0 : SUCCESS in 18m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28ed76ba6522470eaea3a6d34fa83443 : SUCCESS in 17m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d94e84b94ac7423483d8bc9fd63ace73 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/770dbd45978b4af5b6037357e78cf92d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea81d9af251e4d0fa226a1fc37cbc29b : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a8e0b2089fe04a60aedf60ea01956bcf : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b263059dfe1464da4753701068849ce : SUCCESS in 36m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4018e37904964cbb982466f8f5a9f8f1 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d0e6c2d263f41a49540c2859a21c99a : SUCCESS in 1h 55m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d86c3b976c274b96bb42bc5b12bed909 : SUCCESS in 2h 25m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bde472f04613494f910511693ad23d67 : SUCCESS in 1h 52m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9e63b3be6bf449ac9c0cdcca2c922495 : SUCCESS in 2h 09m 15s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27c553cb6d2549f6b2f6ef48c8fed25d : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46b32d31fee6414ebbf44abaae232c02 : SUCCESS in 1h 29m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24ff37e5b8f34108b06ebc0326909360 : SUCCESS in 41m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2418366814a44a7686ba91cf65df49ad : SUCCESS in 1h 23m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/feb1d0ea87c04034b88dee08f77f7274 : SUCCESS in 1h 21m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce16d74354a24a969dcfcbfc550dc0f8 : SUCCESS in 1h 33m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3858385575047a19757e66602f0d1e5 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":2},{"id":"32912719e9fdb3bfa96c7ea7a178924a709fcbc8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-03 15:25:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_07_2024_15_19_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"a60e9e255906c31305593338f762fd4bfb9ae014","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-03 16:41:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/923246/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/f978286/ : FAILURE in 45m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/923246/2/upstream-check/cinder-tempest-plugin-ontap-nfs/d1085e4/ : SUCCESS in 1h 15m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"51b87c2ff6ba6c5360ab33e182299c6f5f9768fa","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-03 20:46:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a418d4587aa41d20cbc64494f7d8165e992dd19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:47:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"692c6c98ade245e154251a5ba577c7d4085fe92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 04:52:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b0c2e99ff34186bf9c56c7a9aeb30d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb6bcee110a47d486ff8f9777666204 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e3ce4a318b4369a5c9f68cedd0c0df : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26e39f46478463b9980cbe712c34306 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6eb7bb66ab3417b960643b7686f2e72 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1532c80574114e0ebcadf0f963ddd870 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a03e293693c640f8a21ea6dac50e8dd1 : SUCCESS in 58m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af5f8010e2014d74a781792e4690f918 : FAILURE in 1h 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6914b92df0874356a81985e0d9a668e0 : SUCCESS in 1h 04m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b2de32414d453089ca9fb3f741e4e8 : SUCCESS in 18m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff473a95fd444811841639bb191e4cf8 : SUCCESS in 1h 14m 38s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/25d8fca9f5274a7eacbb2ffbac80c041 : SUCCESS in 1h 44m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9baa49912c5c4af39a27df74fb24f7cc : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"194fb65b5313e730cd2b71ac3302e3273f516560","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-04 05:34:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6dbf68a8284950df10b467c68446f94068db2a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 05:47:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f5e6e0cf2a42ff91bb870d44e0cd23\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/021682ece2a1487ca57741484a31e6f4 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/019d8fabcdd644a892e55816dea81a3f : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8e43d907975433999519e47121ac226 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21eb9c26a36641769d545e77963f9f2c : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59d499638e8f5d300dacc77de8070ce81c3ff969","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 05:59:34.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/772992a6d1cf472db652d7c68b2b8351\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/69e2a583282a47069688ff38a9dc2015 : POST_FAILURE in 20m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ad312b3a1eab45ed1b38d41d90a9345a7faabd10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 08:52:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09fa05a929c74d55b0c7d3969ffc8397\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40da9b422e94913bea694f1996954f7 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ed40181ddbb465aa2228e28a1411ce2 : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/381c958f618543fcb3240bd9b80b9144 : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89d93684473c4fe9ab2505e8bbf7e221 : SUCCESS in 11m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9b8d0b4feb84308a9ca0b0363e947d2 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44dc98a0ab74d65b68fe6d5d3b17c3c : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/8c6cf8b108be44cc94b269da167df2d0 : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b484255b7c114d82b55b3ca15a5987ac : SUCCESS in 1h 19m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c96b6df3937742ba90fd0c697b8de786 : SUCCESS in 58m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a426ae6f4e0442ba0248ac5cc663f5c : SUCCESS in 17m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a9dc8d410cde42fd801b1182fbb6a8a1 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48f132686ad0479b9ef40622ccf6c55c : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7e29df2f037423da09af62abff3a3ef : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2989a1d607954cbaadf62b7cac13e99a : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa72d5da4cd5486c9af4cff9f1bf4f38 : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea5babe466b4485182a52c7bc3de41ef : SUCCESS in 43m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/363b2eff762b41a086a8aade32ea8b3d : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15e399d6691841ce82a634ac1b699c61 : SUCCESS in 1h 49m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40cc9b580dcf4c7f853a53bdb07ae0ea : TIMED_OUT in 3h 06m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f43e7224e22443e6aa9c523301782b8b : SUCCESS in 2h 01m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e002c779b6fb40c4aeda8a3f857e63c7 : FAILURE in 1h 49m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6c06d922fa3b4d45821612282676f045 : TIMED_OUT in 3h 02m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/988a73581a65483ab19f03cc08dcda22 : SUCCESS in 1h 32m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c919d6f7d3d4c9f8ce842fba1fb378f : SUCCESS in 43m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/58cc2643713d4076a643ad754bf16cf4 : SUCCESS in 1h 11m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/108f7c74daed42aa9c5a93132673e1f7 : SUCCESS in 1h 27m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5cd04a9e40f342e0970917546c8fb1e4 : SUCCESS in 1h 31m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96908dff1ef746c1bc1ef82830bb88ca : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0c679fbde08ebd3cc9911381f7e9acfc08047c4a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 09:45:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6f71f056995d68b3dc634353a9580b0e7f973b6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 10:10:22.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9a4db85c0dab4c0c8049bffac6189292\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a0f99019688d4e7fba4cee99e2ed012a : POST_FAILURE in 20m 01s","accounts_in_message":[],"_revision_number":2},{"id":"355384d12cc5447b815b1d986d70d9c01ef9181e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:14:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19fe5cd1671c4330b635580244ad2f57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4ef652c2cda45699a67bc7ab35c060d : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06504178e6144ec594fa73eabb5dd7e4 : FAILURE in 3m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3cf08657b2d9485487f739fdf3936541 : SUCCESS in 22m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9f40117aa1a4e2c9148b0ad6c6f4006 : FAILURE in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7290cd33d3bc30d7999f8040de132ba6f1fa2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 14:00:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62fa7f8ab8ec430e958a6a26f2f82dca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76854f593b84f049971a7527810fa95 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ad2c41d2834ce88c10096fc61c5923 : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e11e8b0ba5743f886c0f98d8c9f1d1e : SUCCESS in 13m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/594a2cd9e1d841bc85ab5c9943b11f04 : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ad388a142874e2dad736056f9773a46 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba3c9ad18a54fff9328880b6f75fb2d : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/558798df72c346d180fd5e90a2036d15 : SUCCESS in 53m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0154e47952746b9816d75bbf9abe04f : SUCCESS in 1h 21m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e24b6300b0448b193263b962dec324f : SUCCESS in 1h 02m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e940d5bc5c5345369e29e1c24f0070bf : SUCCESS in 17m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2cbe93ee2d0b4de2b0eb3242030e5e5a : SUCCESS in 17m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d38be4ba7a245c48524e53c3e58d211 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8289f5e54c1c4d0eab770f2ac5d60d4a : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a97f11f3d7684d62ab14736f656f5b07 : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/469e904d198c4b91a753696d26ce9b3d : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f5dca2f75624f82a44d55451c3c6ef0 : SUCCESS in 41m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f086a9fa15f44798b8d6d7804f4bf62c : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29c30fdd07864655bf89d12fa7c65a38 : SUCCESS in 1h 48m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b8fe95db5b5d4c1ab55a0004e6b43f31 : SUCCESS in 2h 00m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b48b6e8514dc423888e9b4e8da1c6925 : SUCCESS in 1h 45m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bcee1ddcc0544443be301b0d5915ccc0 : SUCCESS in 1h 58m 18s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8befd5921fe24e1e9833bfceed708f81 : SUCCESS in 1h 13m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e75baf98c20469ebe6ac7a8654862a9 : SUCCESS in 1h 33m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75266336247f484892733d23dfbb6991 : FAILURE in 56m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1365fe5dbc044c61848415f834c6dfc9 : SUCCESS in 1h 16m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0b342b2d5d69447b87c415516c965d13 : SUCCESS in 1h 36m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39b6395f88e541ceac005c2cb8d8a64f : TIMED_OUT in 3h 02m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf9db6961a3e46cca5d0c8c6dc328dd0 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":2},{"id":"536e8be20620d35d1f4dfd41922b18e47edfef31","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-04 16:37:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30e04e1c0e7744ad93aa381c1ce09f8c346aebc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:55:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53567ed4b744185a3c5bfe5557fe9f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7793f1127b04a8d8219a711eec8e848 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64f4af7cf6bf4bbfb80fd66288f3175d : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c5a9e5dcf5c40fca9dc26c5398d927d : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/29a3a3b2205f4358926edb3568afb660 : SUCCESS in 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"536408bf709fecd59515393a40db8e5cd049d165","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 17:00:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/99065f0a63d44024b12763eebdcbb19f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/02a4e874842b40088c2cb9c83594f1c3 : POST_FAILURE in 20m 20s","accounts_in_message":[],"_revision_number":2},{"id":"84c66defa4b2e5fb45131c12026056cb405c5ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:00:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffaf15c2eae04416a7489bc376712fbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca6740828074d68933490dc7bd4f86a : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/390264a4f8ff4f95b21134a165436c7a : SUCCESS in 9m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/717a68e31bae4c9288357920a11ea223 : SUCCESS in 10m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f087a08628a4ed6babf4b064feacb1e : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/832645bcec724598a60026e8c431084b : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217baed00b3449549869856dcb1192e2 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/9217e44b124a4216831c923e415c82e4 : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de528a22aaa74eabbbf6270b4cf821c8 : SUCCESS in 1h 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85441a7cf4474e548dc9846706c5d6b1 : SUCCESS in 1h 02m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85045c11aa904b3195a0c62483afdd63 : SUCCESS in 15m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8a2a40ff36a413bb2b21016382be6ac : SUCCESS in 17m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7050d056139d4302b550990775c02802 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b562a298c43458893bdee5a0086c893 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8a09fde15e542518f325466a22654d2 : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb53cf188e7d4af38e58c73ffc16caee : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5251e7fd262435e849f6946c310c7d3 : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1648e1ab47364de6977fdb02a1a6d983 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e82d0e464b594a3f9a5fdff72bc74df5 : SUCCESS in 1h 55m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/77d91a3237d549aa8e859752b66a254a : FAILURE in 2h 16m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20fc1e6560334eea80665b2f54dcd024 : SUCCESS in 1h 47m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f9e7ba0d39744561acee095e73df3d45 : SUCCESS in 1h 50m 23s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/94a4648fd83d4fc3b20dccdaa83bb305 : SUCCESS in 1h 11m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a22c90c5b7534030a2a1a4ba851cf9da : SUCCESS in 1h 29m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/41f8145b7f5c4a40afd10528c58c3452 : SUCCESS in 48m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dd3064e92422489e85c57efe9a1c4564 : SUCCESS in 1h 14m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/376c40d1fb7e470e878ec94e9a255b7d : SUCCESS in 1h 16m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/617ca61828a942bcae11605676fdc5f9 : SUCCESS in 1h 37m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8afa36fcf1d04667a76b0974663b4ec7 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c05c5dbdc49acdee34b91bc0eae34f695f100cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:19:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"edb46cff57aeea101bdf9399cf3bb619c7a5274e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-04 20:32:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/923246/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/c5138cd/ : SUCCESS in 43m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/923246/2/upstream-check/cinder-tempest-plugin-ontap-nfs/767083d/ : SUCCESS in 1h 15m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f55535c1afc2403a90e67e6d289e69a05031abd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:36:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0da2c81707451094713cf38fb2784c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8816d0fd83c4903868305954afa14f0 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4de8b0be5b524f2b8ee093ae5d39dda9 : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b93faddce0947d19b92dc888b43b62f : SUCCESS in 11m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/370db990286e4f42b17bc206a3904a85 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bd00f9d0564d688f177f501b09e53a : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bd607f82a5024e5cba5e388dc9be302f : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55a9242efaac445fa61470e683010e11 : SUCCESS in 1h 22m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f0259741f204900bed2ace0bc47f91a : SUCCESS in 59m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08c0daccbee14811a049a3073f159170 : SUCCESS in 17m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/70d53d87953a4f079383e693453e3428 : SUCCESS in 1h 13m 25s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71765c595dbc43159ea29adc4841af4b : SUCCESS in 1h 53m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f769112ef0204b7a9e2af3e8364b4808 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b03bf3e79f6d988ab18e8342296808101c8821d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:36:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1f5d11a4abd3a56aefb03c6a223bcb19308a6409","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:39:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b3ba76f2554739b673f429d34a317b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1634c5b108d4beb90b505b2077c7b75 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c99661cf8f5e436db01ba64cc512258d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2},{"id":"00285e4e90ee563a6e56a8be619c818b812738a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 07:30:22.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/184/.","accounts_in_message":[],"_revision_number":2},{"id":"133aa4ede0b97e9681504f87e351e60b7cb8c684","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-05 10:06:01.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/186/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e667b02b2c20b4ada18c1a472be152956284d45","revisions":{"c7fbbba552ea8080965fccbabb4f54855651e7b9":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:04:08.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/46/923246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/923246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/923246/1"}}},"commit":{"parents":[{"commit":"4e00f1b2c1ebf8c819a90c254f62f2552919da7b","subject":"Merge \"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e00f1b2c1ebf8c819a90c254f62f2552919da7b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 18:09:30.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-28 17:45:19.000000000","tz":-240},"subject":"CVE-2024-32498: Check for external qcow2 data file","message":"CVE-2024-32498: Check for external qcow2 data file\n\nAdds code to image_utils to check for a qcow2 external data\nfile, a recent feature of qemu which we do not support and\nwhich can be used maliciously.\n\nAdvice from the qemu-img community is that it is dangerous\nto call qemu-img info on untrusted files, so we copy over\nthe format_inspector module from Glance.  This performs basic\nanalysis on the image data file so we can detect problematic\nimages before we call qemu-img info to get all the image\nattributes.  It is expected that this code will eventually be\nadded to oslo so it can be consumed by Glance, Cinder, and\nNova.\n\nBecause cinder itself may create qcow2 format images with a\nbacking file in nfs-based backends, the glance format_inspector\nhas been modified to optionally allow such files.  Since we are\nmonkeying with the format_inspector code, we also copy over\nits unit tests to prevent regressions and to add tests for the\nchanged code.\n\nIncludes an additional fix to prevent an issue where a user\ncould mount a raw volume and write a qcow2 header with a larger\nvirtual size on it. On reattaching the volume it would have the\nnew larger virtual size avaialable without actually changing\nthe size value in cinder.  While we cannot prevent this we can\nprevent the user from using this volume again, which makes this\nexploit pointless.\n\nCo-authored-by: Dan Smith \u003cdansmith@redhat.com\u003e\nCo-authored-by: Felix Huettner \u003cfelix.huettner@mail.schwarz\u003e\n\nChange-Id: I65857288b797cde573e7443ac6e7e6f57fedde01\nCloses-bug: #2059809\n(cherry picked from commit b207be3278f3dab6cfe945bd3d307db168f6f6c7)\n(cherry picked from commit fe0940994ca93344c4d4a9923471ada6fb3a38ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7fbbba552ea8080965fccbabb4f54855651e7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7fbbba552ea8080965fccbabb4f54855651e7b9"}]},"branch":"refs/heads/stable/2023.2"},"9e667b02b2c20b4ada18c1a472be152956284d45":{"kind":"REWORK","_number":2,"created":"2024-07-03 09:15:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/46/923246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/923246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/923246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/923246/2"}}},"commit":{"parents":[{"commit":"4e00f1b2c1ebf8c819a90c254f62f2552919da7b","subject":"Merge \"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e00f1b2c1ebf8c819a90c254f62f2552919da7b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 18:09:30.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-03 09:15:25.000000000","tz":330},"subject":"CVE-2024-32498: Check for external qcow2 data file","message":"CVE-2024-32498: Check for external qcow2 data file\n\nAdds code to image_utils to check for a qcow2 external data\nfile, a recent feature of qemu which we do not support and\nwhich can be used maliciously.\n\nAdvice from the qemu-img community is that it is dangerous\nto call qemu-img info on untrusted files, so we copy over\nthe format_inspector module from Glance.  This performs basic\nanalysis on the image data file so we can detect problematic\nimages before we call qemu-img info to get all the image\nattributes.  It is expected that this code will eventually be\nadded to oslo so it can be consumed by Glance, Cinder, and\nNova.\n\nBecause cinder itself may create qcow2 format images with a\nbacking file in nfs-based backends, the glance format_inspector\nhas been modified to optionally allow such files.  Since we are\nmonkeying with the format_inspector code, we also copy over\nits unit tests to prevent regressions and to add tests for the\nchanged code.\n\nIncludes an additional fix to prevent an issue where a user\ncould mount a raw volume and write a qcow2 header with a larger\nvirtual size on it. On reattaching the volume it would have the\nnew larger virtual size avaialable without actually changing\nthe size value in cinder.  While we cannot prevent this we can\nprevent the user from using this volume again, which makes this\nexploit pointless.\n\nCo-authored-by: Dan Smith \u003cdansmith@redhat.com\u003e\nCo-authored-by: Felix Huettner \u003cfelix.huettner@mail.schwarz\u003e\n\nChange-Id: I65857288b797cde573e7443ac6e7e6f57fedde01\nCloses-bug: #2059809\n(cherry picked from commit d6a186945e03649343af55b46ed8dfe0dd326e40)\n(cherry picked from commit db98dc207060da234c32a563c13cac1edbd62952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e667b02b2c20b4ada18c1a472be152956284d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e667b02b2c20b4ada18c1a472be152956284d45"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
