)]}'
{"id":"openstack%2Fcinder~923314","triplet_id":"openstack%2Fcinder~unmaintained%2Fzed~I65857288b797cde573e7443ac6e7e6f57fedde01","project":"openstack/cinder","branch":"unmaintained/zed","topic":"bug/2059809","attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-08-12 10:42:54.000000000","reason":"\u003cGERRIT_ACCOUNT_37205\u003e replied on the change","reason_account":{"_account_id":37205,"name":"Dmitrijs Fedorovs","display_name":"Dmitrijs Fedorovs","email":"dmitry.j.fedorov@gmail.com","username":"df704292"}}},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-07-12 21:19:09.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-12 21:19:09.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-11 14:48:17.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I65857288b797cde573e7443ac6e7e6f57fedde01","subject":"CVE-2024-32498: Check for external qcow2 data file","status":"MERGED","created":"2024-07-02 15:36:55.000000000","updated":"2024-08-12 10:42:54.000000000","submitted":"2024-07-12 21:19:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"923314-bug/2059809","meta_rev_id":"73877326301f612c13d48aefaa7e3f3e2363118f","_number":923314,"virtual_id_number":923314,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-12 21:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2024-07-11 15:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-07-11 14:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-11 13:30:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-07-11 15:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}],"CC":[{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":37205,"name":"Dmitrijs Fedorovs","display_name":"Dmitrijs Fedorovs","email":"dmitry.j.fedorov@gmail.com","username":"df704292"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 15:36:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-07-02 15:38:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 15:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 17:39:11.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-02 19:05:55.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-02 19:09:46.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-02 20:32:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 05:36:32.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-07-08 15:52:04.000000000","updated_by":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"reviewer":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"state":"CC"},{"updated":"2024-07-08 16:36:48.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-10 10:50:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-07-10 13:15:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-07-10 15:13:02.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-12 10:42:54.000000000","updated_by":{"_account_id":37205,"name":"Dmitrijs Fedorovs","display_name":"Dmitrijs Fedorovs","email":"dmitry.j.fedorov@gmail.com","username":"df704292"},"reviewer":{"_account_id":37205,"name":"Dmitrijs Fedorovs","display_name":"Dmitrijs Fedorovs","email":"dmitry.j.fedorov@gmail.com","username":"df704292"},"state":"CC"}],"messages":[{"id":"b656ba102cfc84f6b2f2cf54f32605e64ec39762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-02 15:36:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62ff9cfb2039add9909d8162f076303d724ebad4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-02 15:38:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"dbe7971e1e2a7846b83cc8f04c7c9785dba64d61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:41:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a40d6a5b981c41dab6ef6d4a3597cde8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/987ed740193443ae8b807e9bf257c047 : FAILURE in 3m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b866aa918599455f9db5072426fd3605 : FAILURE in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cdaee8477cdf86886c10f10d4904dc5eec7cd22","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-02 17:39:11.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b0431134ff6c4f83b3ee25713a51e7c9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f9c4c25344cf44d39e46f2ba8c7a6dc3 : POST_FAILURE in 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"22eae3f9cacac55b7824a9ee58e481ee379af25f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-02 19:05:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923314 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e0a2f4bd18756d7a59979d46fc886af1fdbf5c2e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-02 19:09:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923314 : FAILURE in 40m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fec5154f8fedf77e1d6b20d47fc3f6cf1d2237ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:32:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c035ec35ddb441098a8f5450f60f1d86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5864d9bd06ce41f1a25ee1b74d5bef4e : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a11e45cf40d43b5a617d985738f6ea9 : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccde8f6da2254c70a3a8acd98b6700df : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13a168bc79ac4339b61e8ad76e045296 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4095f32a3b4f0981a233178c4f6b31 : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce7111cefba24d1694dccd6fc5d341a4 : SUCCESS in 17m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6a7fe3871784498b0eafa650293528a : SUCCESS in 14m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f710480348f4f708699f0b341f98da4 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed0ffd4783e3459395866e989296b1cd : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9af78fa48ab74a229c71b52c9a4e6920 : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/199f9287391a4035abf9069b322781a8 : SUCCESS in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e45e2c5eebab48eaa4f83f9eb4a3f904 : SUCCESS in 42m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a26ea87f5b049fcb7670b1809e0c0ca : FAILURE in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c356f1a8e93c44d9923370715b9fa5e5 : FAILURE in 1h 55m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/24925f1d82054b06b103aa36389c2606 : FAILURE in 1h 20m 39s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2708911d02b4454981acf099640cc2e7 : FAILURE in 1h 14m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f5248b6941b54f2a86863624766dea4a : FAILURE in 46m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/028d48e222c2403f8d38438fe9858557 : SUCCESS in 1h 17m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11c7075fee114ce2b67aaf20630bfdd1 : SUCCESS in 1h 10m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/335b91b95ae847fc8207d429983ee91b : SUCCESS in 1h 32m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd378d62c6ad4c698d9aabd8bab83092 : SUCCESS in 1h 01m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc57a772dbc6426e904489a07a7f1328 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"05d3d9844bb728ae4d5fc18347c179b3ce9ec394","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-04 05:36:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"923fbc8e1185e4bd6151897fef640e378907d264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 06:02:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e96f441f5cc4f419406969dfe3f0ae9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccaa8496870648f99173d2ce1c9c34c0 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac538f2116454e59bd40a9140e6074ca : SUCCESS in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b429f55b9c4d318c3144c38c1fd45fc56d92e20","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 06:21:59.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/785f65d90d41475f82783496b4e09eb2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/22c11a97a2ea4be7a1db5605efebab1e : POST_FAILURE in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"28ef5db7c0e1ed2baf5ba181db403ac1d3bd304b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 07:19:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7747068aae742bb9a03635808f58526\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00332458340a4cc18dc6764c1e2b72ef : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5188542bf9b4147b86a12b1eeeacf2f : SUCCESS in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d80c333f772471c9a9f62a982d29e89 : SUCCESS in 12m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef69cce21008499d888d654ed05c49cd : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc26a6ceefa49239ecc9ae69ca87303 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b5e031054404b13adb562e1a7c29bde : SUCCESS in 17m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7143dc243884c40acc50c15fb4ef1c8 : SUCCESS in 18m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3da7366884b248ddb6455ceb89f167c8 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36bbf1fbf32b465b8c48c50564567d7d : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/533feb5db578489dbccbb2a353945eb7 : SUCCESS in 5m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e66a7a3e951448278edde547108ad87d : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a62cd63488f24fbc9ee96344251c574f : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/973bf1d735a7431cb23514c30206364c : FAILURE in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8697640348547a4b8ca43423e5e3854 : FAILURE in 1h 34m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a83c3d17fcee4bdc8666d01e5448d5ac : FAILURE in 1h 32m 09s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2102b74714834efda8778b2665516844 : FAILURE in 1h 08m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38d0f958b93d470da89334a39677c86f : FAILURE in 40m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/266bafd6d1f34f29b85e15742066034d : SUCCESS in 1h 07m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45f1ee8234cf45f0a1d7f7ded478c93b : SUCCESS in 1h 18m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e6e9c4b16de42c0afa653206386d0d9 : SUCCESS in 1h 01m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/010459875c934656b5302f22abad0c39 : SUCCESS in 45m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0dc3cf6a38c4cfe817a31dd596d66a3 : SUCCESS in 47m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d935ae648abbe0f27df4e5a95079a61e1ea91332","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-04 10:05:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"932a8572d57c23236c12b90f013413c4f5a01e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:30:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/828c912c977a48858ecd443c7cbc4792\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e4ac38c2b9d4fe3b8ba98f513d2dab6 : FAILURE in 3m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c31281b5b654275bf17c379ab90af7d : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9fdc69ff4f98f91104de5f69df9e823a9ba9c72","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-04 10:31:13.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9bb20835c5d74a219bb1da5f920e3045\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a3a29a3d78c2443f89ccf52187f65023 : POST_FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b4b5f533291f18bae82e2d70e9195440ba4591f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:20:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89a96f0c2b414f53bb1755aa5e04ad79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93cfe391c8549b38a8483458dc720a7 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e4ca2b51df14bd0bdd299fcf15aacc4 : SUCCESS in 14m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d49710c7a0a4cb0897f4d9b34652783 : SUCCESS in 9m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b17f9823a62445afb40ea0f0880a694d : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5901ccc5cc984e87b8c145450874be25 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643672d59aea4d98a5f8990ae92e6480 : SUCCESS in 15m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70f450e9ce0141849cfbd74efb2234fd : SUCCESS in 15m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/89126a7bcfc1472f991e5d18ded4d530 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0253d2cb41c448e7a6f9a61af2b9161e : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7d4585a876d4f06801f3be0497dc8d1 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f7c29d5e9ad2404a93b43ad21aa4cdee : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e5535b1cb33d4bf08f6fd934dbafa3c1 : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f96d49c8ee0a4ae4b0501af54e758be9 : FAILURE in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f85f24aca6a24b908ab5d929d6d74e78 : FAILURE in 1h 34m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/56a5977ef9254dd294337cd34e310135 : FAILURE in 2h 56m 52s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/609ece72108e41808ae967ff0ff826a2 : FAILURE in 1h 09m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/55660cce19de45589bb5ea9adad7d02c : FAILURE in 39m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f9a9f810a194b24a77f736766b9c39d : SUCCESS in 1h 05m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/539068ad70314e7997edd06b3184ebeb : SUCCESS in 1h 23m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44924c46584f470d928dfc66264398f6 : SUCCESS in 1h 06m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13f3d92ed8814820bd630b815c055f46 : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/479bcef84c034ce7a4270a0b507b6e9b : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"604be9c492d9eec2943988c1f67d0d87f6afe967","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-04 22:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ef5daac0f5165cd5a88f4617ce220426dd7ba7e","author":{"_account_id":1313,"name":"yaguang","display_name":"Yaguang Tang","email":"heut2008@gmail.com","username":"heut2008"},"date":"2024-07-08 15:52:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28d16a919dcd2ccd909641e26cad53aa3d8af0ed","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-08 16:15:04.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/209dbca29ac0440594cee0b79852b7e2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a158679d470f4677a998080f995af9c6 : POST_FAILURE in 20m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a4ba5dd67c3cd0dff3f1806db6b9f6ce7ff9961e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 16:19:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b18099f114d4eb58c7a1543641f4bc0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9468edefd2484b42bc723e0de9e3ff6d : SUCCESS in 26m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fca3bd58a9b47bf85ee9ecbd5edfa65 : SUCCESS in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50667c6fdd6f8840ad12fc40f8d847ddc5c39d01","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-08 16:36:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c076a0aa1f0344bf7e92e5c6f582ca7433956b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 17:37:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84741a1441da4a848e10648438fe2beb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc6f30306da48f0afbf3b7c34c4997c : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9305cd8583bf41f7956755f2a7692273 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89a60ee853144e16bec2f282a1269e46 : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61979ecb8c21420381ff29763e1a62e6 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fdbf7f5fc944769f4b3ed2e989b4b9 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26a3ef62a09645988227788e066a04aa : SUCCESS in 16m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb7b361ab7244189ba3091c8d87d3b65 : SUCCESS in 14m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/58ac680c106c46c1972faef6005085e2 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9fa5cd4679d74a64b80061be5317b1cf : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/278bdcb362eb427e836d2cd384cf25d1 : SUCCESS in 4m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f946ae8afc4647b983fd955a9b7bffec : SUCCESS in 5m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/abe5610512e64fc1971596ba733dce1e : SUCCESS in 28m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92d6867dd3834846abcba835bc67073c : FAILURE in 4m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d2c2447a70f47a1bf4323cd39284a9a : FAILURE in 1h 44m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a3f03753135841fb98245fcc27c59f13 : FAILURE in 1h 39m 08s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6581fd6b7c354da5a3cb345f5ac699ba : FAILURE in 1h 25m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8902849992b4d99a8634fa3145a88d6 : FAILURE in 43m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d823d9cbf889438b92c33dec26b3880a : SUCCESS in 1h 02m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47812c24dc9e4f2390a6c857a9fef575 : SUCCESS in 1h 15m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13ccbab619d9493c93a66195c0aa2a41 : SUCCESS in 1h 05m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b2da7f6c12a411590dbe11d1fef64dd : SUCCESS in 57m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45de3760900d4e5aa44e52a1de127ef5 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"921a418d758f2970e002ca9624af309841053d6f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 10:50:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d82f26dc3e119e10481cf6bbb86f579bbb4da94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 10:54:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72854354534d4a51beee52d8b7136339\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbde0eac41624c4192561ba8e3c8b253 : FAILURE in 2m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0264861ad12e486ab06821e6de26b47a : FAILURE in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8978ccee4295dc41c2084ce1cc07eab323d03ab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-10 11:14:22.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b6497fc5ed14423f89fc37f8b618414f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7c72a3a7b3ca4d88a39b62884d15b02d : POST_FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"10d9ffce4ec6f50a7a4eb8e1cec32285b827b898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 12:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc09425e9fd4005910a63221f2a1cde\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be49351f285c4f6688b9e8eee7359dc2 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebb6ad41f77b4cb684622f8ae0207b19 : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/debbb97434ff4334978b7f1746c1deaf : SUCCESS in 10m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1aaec2260c9544bcbe9caa90c14edf38 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a52f81705364d6381e0e8168e260af6 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b2f17fe31e2481cbb5bc3ccbb6fda91 : SUCCESS in 14m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1891fb56346341cf9d916f6839fd74eb : SUCCESS in 15m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d49687602d1d4d93b2008e251aa2b24c : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a7c5e98495343179753976ba80c7847 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3079c853e4f742d589efc1e487836545 : SUCCESS in 5m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9e2e3c29f654555b66be3d534eb2db4 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a18de8b88b2347788a1cf143b95e8def : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c7d68fef242446eaba40ec289691441 : FAILURE in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/049b1832d030473ab9efce84fdb54b94 : SUCCESS in 1h 47m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57b4abd8de8c4763a0b522c05d5b0700 : FAILURE in 1h 26m 09s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6eebe93b12174fefa79138645aa368c2 : FAILURE in 1h 36m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8c78dbcfede4950841d921357da5add : FAILURE in 45m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b719523939734ab8bc8336bc5081d6a1 : SUCCESS in 1h 18m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e392751da0841679aac681180b78196 : SUCCESS in 1h 18m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e946f7050e12446985a269814909287b : SUCCESS in 1h 21m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe2408b4a97a4bde82f7be86b624172e : SUCCESS in 52m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/610fff54665844c5b91b84ac8a3c0b1a : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1826a20c3ac69b5aecb01b2bec543afaeae0b0a5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 13:15:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3f4a23a4a1cb9e873104516851e241f6cb11ee02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 13:16:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"52f0717245f502b9c228312c2c4a5adb04ff467e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 13:20:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c06c3fc70874344b66e1062ad0be4bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7aba147ebc24536995b2c8e4b2da8f1 : FAILURE in 3m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f44edc03e484a439bd89552355b082a : FAILURE in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b56fbd774334c7d400047f2b08f9d606d67f28ba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 13:21:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ede2823136ef8af54c7094baa9d2879ffe46947","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 13:24:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ca0ff01cec984392365c799f0576d8a6eadcd87","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-10 13:38:32.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/53dd8c0778a1402b9f200215231f83d5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/19c4cae5ef4f42fa9ba8807965855a7c : POST_FAILURE in 19m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c85ee76f12f31783c8924738990ad5088f67e7c8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-10 13:49:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923314 : FAILURE in 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ee088bb4ce3ca930a5930533d8ddb1cc11ee06e4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-10 13:52:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/923314 : FAILURE in 34m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a7179f65ff2c29e317cefd040b9930918c1f1cdc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-10 13:52:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923314 : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f2c906dffcc393d794b915e89f163f8e0ad48fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 15:02:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81dc74359a704dbdbc0bce1b2403cd2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ffcf1b99294f1c84c2413d7a93265f : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98ed968c3a8c46038f0ae5afcb4ae9a5 : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2935588d6d2241ceb111b9e696a69259 : SUCCESS in 10m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce36923645b3429fbfea280130760584 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ca8834bb8d49a7a862467967bc03c7 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fa3d37a84374eb9a5bb46a654ec312d : SUCCESS in 18m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4fbd3c1e3a9a472090150d68b9ea9ef4 : SUCCESS in 14m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e08b51846ce40baa2b049bf116fc4ef : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a68954ccfb81480fba73372cc64f056e : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b3a36fea88746619893cc361110d78a : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b855262b38214528be629b140f162b83 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5ba94751a164035a44a781b85429d2d : SUCCESS in 41m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13b1abd458414e02a685493ef76ec84f : FAILURE in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe67329b7a524a7aaa01ab37ffdd4631 : SUCCESS in 1h 38m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e26cd911368c462d8e53b9276b599d3d : FAILURE in 1h 38m 16s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72546a69fe854b3888d35f86c9976cd9 : SUCCESS in 1h 32m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bfa5c66189bc4489b22821a348871969 : SUCCESS in 35m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45140cd2017a4b978d21a9e736787e50 : SUCCESS in 1h 05m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e8340e00edd4ea7bd17e036d11fe6c6 : SUCCESS in 1h 20m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3296eb74757472e8b59e4450ce609c3 : SUCCESS in 1h 18m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a035f38bfd34520b6d5a5bf97ffe930 : SUCCESS in 53m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e4a0b7e84d14b31833436fcac8a75de : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4560cc3dc992f8491641dba7fd832825814542d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-10 15:13:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/923314/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/cbce1a1/ : RETRY_LIMIT in 8m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/923314/2/upstream-check/cinder-tempest-plugin-ontap-nfs/e3e190a/ : RETRY_LIMIT in 8m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b65f6ac91ab0bc88344a47049e2d99dd18de9ac","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-11 13:30:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6567cfe9f18a4060749a72af26faf334282a650","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-11 14:48:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bc00893f914110e5c62e88c74c8507f8e9f1dc57","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-11 15:41:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3636a1136b83b67c5dc0deda58014fe1a182aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 15:42:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8230e52936f53385693ce7e1bc4fbad3d616c1fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 19:32:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a8b4459a661467ebdddd7e976e89d40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10df309df4414289b0c48352c1345fc8 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3898a93ab7547a8aab7bec55f7fadd3 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7659b4fc2ed436a81531aadaa89aa86 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ea6aa34b904775853c39c993b936f9 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff88db9c3a1c4231892cc008b6dcd4ac : SUCCESS in 13m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ceb0b4c1b04846859e34f2e65f264c96 : TIMED_OUT in 2h 05m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/767f9db047014970816b8ae64edc8439 : SUCCESS in 1h 03m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c85cb79ff658434197e7c211e694ed10 : SUCCESS in 54m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/617dcd75ec36471585aa894b1e910ac3 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":2},{"id":"30f4bb4722a81d340bfd1ffa97a84611dfb78e25","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-12 06:44:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f880ab5c23e67ce109f8ddc358c06716d9d75638","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:22:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203b627fcdf149f3891def15cdd26674\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8aa3c3924734b68a2943189c63978c4 : FAILURE in 32m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c811d93ba47a4b0cbb3a4671d1424d34 : FAILURE in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed303ccd46cce5e6993dc2e90355a854e58162f3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 07:39:01.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cf88158330844960a24365ac6e0284bc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/96a3cd4dfaf545cb8b62d942facdfc73 : POST_FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1080c24cfe9def38ccb3e5d450bb576d4737c694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 12:29:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ea738b8321b405ea15aac4fae88eba2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb1c1f9e6e94cc4a868f97d7152120f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37870d154a2145fe9960d8008a2f4a6a : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df83a239518047998ce810048ab63070 : SUCCESS in 13m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b27094733bab48b39c0f50c5485b76c2 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6b1c3636244633b7d3edc91fa8f2e0 : SUCCESS in 13m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4185c464cd5e4c168191e399117faf2a : SUCCESS in 18m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/793fa995cdaf4a25accaa2a782c50c31 : SUCCESS in 14m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5a45e8d9b69455aaff5b7f1cb87e0ae : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b0c9be2d7b04d458504c7181e5bf1b2 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/baedc1f97dbf43cc82936f5d039828d7 : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d947be20975d4133b8a25a2267b2ca25 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ff82f6773d0e4385a5081190d9d665de : SUCCESS in 50m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1cd732b7bd84c778411060ef2a568b6 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/516143b7b1204c40b52b2fbf8f56eb1c : SUCCESS in 1h 35m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e2d7c8655e114f8ca07302abbad21f20 : RETRY_LIMIT Host unreachable in 1h 12m 03s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df27ed9884ff4b289c8dfb716bda2214 : SUCCESS in 1h 23m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c19f21ffa23b4136b4c6f4e51361805c : SUCCESS in 47m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e2cec89d573e4fa987d17752298e9837 : SUCCESS in 1h 21m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4cadeaec2ec74a56b573398545da0689 : SUCCESS in 1h 15m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/641e1aaf10ca4e7195c318fd0c3f2f6e : FAILURE in 1h 11m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfcf99c439de4204a47f2b2dd56be45b : SUCCESS in 53m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0553ced112414767b24f3f8359e0d3ef : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"917f0904e980be82e79e255cd6abb7e17e4633f4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-12 12:48:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d60cbad0dcce50f9c9b7f61bd8e373fb5742de7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 13:00:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd2d959fedf4021be0bda5775100eff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f70c9cd8299142b69a2a2467d03abf47 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ea2cade45c340eab9733fb9fbedbdd2 : SUCCESS in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d9381e8fca6c85c7b6fb3f09eb92b4e49797330","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-12 13:16:59.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f7681f65b5f84eefabf7a1d8b4c4cf84\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/258d73f7942a484e810ccc940bcf7885 : POST_FAILURE in 19m 21s","accounts_in_message":[],"_revision_number":2},{"id":"294bd974808dd9be492551164b5d0352c55bd315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:33:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dda3d96ef0d4f47b4918c1b279b3721\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4213b72dd78145258ead1e493f9234e4 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d1ce0b5c244dbcaa24fcd6003fcd3d : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09988b1180cd429097c909219b18175a : SUCCESS in 13m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db586c178f21467ba124fd02c80d2e9c : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2dc4c7aaca844e5889af43fdd6dd844 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf561df5f99b4f938b591ab3ce3df188 : SUCCESS in 19m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ebb75a8040342c28c8b4ca8a4f635a7 : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5689843ad914c5c836678472126870c : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0646953d2144240a4e81b44aa9dbecd : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9a02b20093a49958a43ad4a97ba7c64 : SUCCESS in 5m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d583b3bd2e44c74a9a4c05bdd203302 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f32cd8d040ed4649a2438d435fb2e141 : SUCCESS in 44m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0014f396d38c43b99186746dd956ae6e : FAILURE in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a76b83d20b814b239f20ce73e715adbf : SUCCESS in 2h 00m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d776ddc3664c469e9620c0aafef52bc8 : RETRY_LIMIT Host unreachable in 1h 13m 32s (non-voting)\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a4f76dc24cfe4b7489f642ac31afdd70 : SUCCESS in 1h 23m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc81a6f76873408585f64f047c6fd6ea : SUCCESS in 45m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ef613678d5045beb9db02e1a842ebfc : SUCCESS in 1h 01m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bd3dfdbc7b7402b82011f0c3aa01eac : SUCCESS in 1h 15m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0cb34e0fd954692b6b83b805bd3f0e9 : SUCCESS in 1h 22m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ae59a56f49a4c17be408d5d595f460e : SUCCESS in 54m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edfdee4df78649539a48a9cc726d6c8c : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6d59ba9e5b1df1cb3c4ff4cdb2f7e27189f92016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 18:33:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf581e107f498f950f112bb2ee5d3a58aa3a9169","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-12 18:48:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/923314/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/b99bf3f/ : RETRY_LIMIT in 8m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/923314/2/upstream-check/cinder-tempest-plugin-ontap-nfs/db050ab/ : RETRY_LIMIT in 8m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2bedd917efa8ae2139077dd80e8cf3a91314d0d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 21:19:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4119d9ae1fb64e27865398188b1172a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff3a38046fd484a99fb79fb7348b798 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef6089631fb6486cae1719f05232d824 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85c5321567234e339601396dd6b35601 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a265de5b12144ac39410c0c748570153 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddac207b43264020b5aa13facc0a548f : SUCCESS in 18m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fb9ec14ad1d74a80a31d0bc6cac8e045 : SUCCESS in 1h 41m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18481cec17ba45538843be02b7cf2221 : SUCCESS in 1h 24m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a47ce71ddefa4d4aae6e4a7ccd37a0bd : SUCCESS in 49m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1a6c2cd85bd4e10b4d0d97a4d8addc9 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4b67b1b56a12aba3c500db2d82c64b542f5aed83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 21:19:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7661e648a9e27f19d392ac3128ad8723006dc8a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 21:20:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9b5d320955476fbe86b51417454141\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c8d557373d4d17aaba2064dc30c7c5 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d8d48263c464cdf9f7455484089fd7d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2},{"id":"73877326301f612c13d48aefaa7e3f3e2363118f","author":{"_account_id":37205,"name":"Dmitrijs Fedorovs","display_name":"Dmitrijs Fedorovs","email":"dmitry.j.fedorov@gmail.com","username":"df704292"},"date":"2024-08-12 10:42:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4aa6590a483901de64e0d162fff11f3d2d7f9977","revisions":{"0a5cbaecd97bae8b9b33b759822288395849995d":{"kind":"REWORK","_number":1,"created":"2024-07-02 15:36:55.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/14/923314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/923314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/923314/1"}}},"commit":{"parents":[{"commit":"30941ad410abd7797b864c5c0ba4f2a3229ec49c","subject":"Update .gitreview for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30941ad410abd7797b864c5c0ba4f2a3229ec49c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 18:09:30.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-02 15:36:43.000000000","tz":-240},"subject":"CVE-2024-32498: Check for external qcow2 data file","message":"CVE-2024-32498: Check for external qcow2 data file\n\nAdds code to image_utils to check for a qcow2 external data\nfile, a recent feature of qemu which we do not support and\nwhich can be used maliciously.\n\nAdvice from the qemu-img community is that it is dangerous\nto call qemu-img info on untrusted files, so we copy over\nthe format_inspector module from Glance.  This performs basic\nanalysis on the image data file so we can detect problematic\nimages before we call qemu-img info to get all the image\nattributes.  It is expected that this code will eventually be\nadded to oslo so it can be consumed by Glance, Cinder, and\nNova.\n\nBecause cinder itself may create qcow2 format images with a\nbacking file in nfs-based backends, the glance format_inspector\nhas been modified to optionally allow such files.  Since we are\nmonkeying with the format_inspector code, we also copy over\nits unit tests to prevent regressions and to add tests for the\nchanged code.\n\nIncludes an additional fix to prevent an issue where a user\ncould mount a raw volume and write a qcow2 header with a larger\nvirtual size on it. On reattaching the volume it would have the\nnew larger virtual size avaialable without actually changing\nthe size value in cinder.  While we cannot prevent this we can\nprevent the user from using this volume again, which makes this\nexploit pointless.\n\nCo-authored-by: Dan Smith \u003cdansmith@redhat.com\u003e\nCo-authored-by: Felix Huettner \u003cfelix.huettner@mail.schwarz\u003e\n\nChange-Id: I65857288b797cde573e7443ac6e7e6f57fedde01\nCloses-bug: #2059809\n(cherry picked from commit b207be3278f3dab6cfe945bd3d307db168f6f6c7)\n(cherry picked from commit fe0940994ca93344c4d4a9923471ada6fb3a38ef)\n(cherry picked from commit c7fbbba552ea8080965fccbabb4f54855651e7b9)\n(cherry picked from commit 96461c2b75a7a7e58b55a1ff6068625096be8064)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a5cbaecd97bae8b9b33b759822288395849995d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a5cbaecd97bae8b9b33b759822288395849995d"}]},"branch":"refs/heads/unmaintained/zed"},"4aa6590a483901de64e0d162fff11f3d2d7f9977":{"kind":"REWORK","_number":2,"created":"2024-07-10 13:16:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/923314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/923314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/923314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/923314/2"}}},"commit":{"parents":[{"commit":"44b9f08ce9d221e3dd0ae17c81b6fc8bf349c940","subject":"Merge \"Update .gitreview for unmaintained/zed\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44b9f08ce9d221e3dd0ae17c81b6fc8bf349c940"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-06-26 18:09:30.000000000","tz":-240},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-07-10 13:16:18.000000000","tz":120},"subject":"CVE-2024-32498: Check for external qcow2 data file","message":"CVE-2024-32498: Check for external qcow2 data file\n\nAdds code to image_utils to check for a qcow2 external data\nfile, a recent feature of qemu which we do not support and\nwhich can be used maliciously.\n\nAdvice from the qemu-img community is that it is dangerous\nto call qemu-img info on untrusted files, so we copy over\nthe format_inspector module from Glance.  This performs basic\nanalysis on the image data file so we can detect problematic\nimages before we call qemu-img info to get all the image\nattributes.  It is expected that this code will eventually be\nadded to oslo so it can be consumed by Glance, Cinder, and\nNova.\n\nBecause cinder itself may create qcow2 format images with a\nbacking file in nfs-based backends, the glance format_inspector\nhas been modified to optionally allow such files.  Since we are\nmonkeying with the format_inspector code, we also copy over\nits unit tests to prevent regressions and to add tests for the\nchanged code.\n\nIncludes an additional fix to prevent an issue where a user\ncould mount a raw volume and write a qcow2 header with a larger\nvirtual size on it. On reattaching the volume it would have the\nnew larger virtual size avaialable without actually changing\nthe size value in cinder.  While we cannot prevent this we can\nprevent the user from using this volume again, which makes this\nexploit pointless.\n\nCo-authored-by: Dan Smith \u003cdansmith@redhat.com\u003e\nCo-authored-by: Felix Huettner \u003cfelix.huettner@mail.schwarz\u003e\n\nChange-Id: I65857288b797cde573e7443ac6e7e6f57fedde01\nCloses-bug: #2059809\n(cherry picked from commit d6a186945e03649343af55b46ed8dfe0dd326e40)\n(cherry picked from commit db98dc207060da234c32a563c13cac1edbd62952)\n(cherry picked from commit 9e667b02b2c20b4ada18c1a472be152956284d45)\n(cherry picked from commit 5f5e86e3542866227b7339713148b5169d069f21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4aa6590a483901de64e0d162fff11f3d2d7f9977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4aa6590a483901de64e0d162fff11f3d2d7f9977"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
