)]}'
{"id":"openstack%2Fcinder~923981","triplet_id":"openstack%2Fcinder~master~I41d706728b2f313378616d005d9069361886f995","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-07-31 20:21:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41d706728b2f313378616d005d9069361886f995","subject":"Tests: Mock sleep in fungible unit tests","status":"MERGED","created":"2024-07-11 17:52:51.000000000","updated":"2026-07-31 20:22:26.000000000","submitted":"2026-07-31 20:21:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923981","meta_rev_id":"7c859c93b1051ae2aa1a680d2fba75f493d10cc7","_number":923981,"virtual_id_number":923981,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 20:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-11 19:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-07-31 19:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":1,"date":"2024-07-15 08:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-07-31 19:45:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 17:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 18:09:49.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-11 18:12:21.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-11 18:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 18:20:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-11 19:16:33.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-11 19:31:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 19:52:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-11 19:59:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-11 20:06:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 20:21:43.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-11 23:26:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-12 14:06:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-15 08:42:28.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"REVIEWER"},{"updated":"2026-07-31 19:45:50.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"16a4bc8b89b71c4b6eb63f4b7d33a14ec0b41e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-11 17:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed8051a86041cab355e40cf92637302ae5a234b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 17:57:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f4e063da9840b9a7bd0680cad87209\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cddf461720a84bea9d1fca3492420d23 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04450f6ad8a6410080cd3996f06ce2d6 : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f107c8cf8a8078dbea31b49b9a8516e2e5d99f38","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-11 18:09:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/923981 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"31959d1d612073af98f1ef995ad4fa36d4d42189","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-11 18:12:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/923981 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4e745e2b79c6dd38e48c02045a106d10653cd29b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 18:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5e4b16c2ad4ed9b7a79a760a1c9e28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad04baaa9d3c48679409f5dea79403db : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30e59d624a564bbb9a2dbc22ac294c7d : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a75dfc971ea24e0f8df8563f7accd2bc : SUCCESS in 12m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/571bb3f387934b909b7a3949edd0373f : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e4b99831614b7db896f419f2d96d21 : SUCCESS in 11m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ad447ea141b467db79b3471acda581a : SUCCESS in 15m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1a32f666bae46b9888e25144eb99191 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8035c5ef08e30225fffe326845b5288cd6641aff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-11 18:20:58.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f898cc37b717403bb9884394e4addec7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/419e5d8b53d645688e0d0d36e53c38a5 : POST_FAILURE in 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6644102fc475c12a1409fce7dd89356b5d27a5ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-11 19:16:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/712/.","accounts_in_message":[],"_revision_number":1},{"id":"b17e7cc632f7d88eaf9aa83c9d2f77189529b1e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-11 19:31:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c2839a6152b941b0a930ea36257105a5 : SUCCESS in 1h 14m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/be7d3c0d62674e2db9a0b9a98a377ac1 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e5d6bb1cc0acdc818980e79682851ec7bb89b300","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-11 19:40:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/215/.","accounts_in_message":[],"_revision_number":1},{"id":"c968c795be56fd6bd7e70b4f69abcdcf7f775d42","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-11 19:52:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"792cfe8dbd251bd64f0671638c5b961bca4fa4ad","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-11 19:59:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_Unity_FC/e6c4afe/DellEMC_Unity_FC/2205 : SUCCESS in 1h 34m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_Unity_iSCSI/8cec07e/DellEMC_Unity_iSCSI/2214 : SUCCESS in 2h 06m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5f2f8611523f770b3251d367d596643b104c1606","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-11 20:06:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerFlex_3.6/3c6a80c/DellEMC_PowerFlex_3.6/1690 : SUCCESS in 2h 13m 11s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerFlex_4.5/d1fbf36/DellEMC_PowerFlex_4.5/571 : SUCCESS in 1h 53m 04s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/b6240b6/DellEMC_PowerFlex_4.x_NVMeTCP/525 : FAILURE in 14m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"966e7d2e5b06b07b0b7d57135921e8b004241be6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-11 20:21:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerStore_iSCSI/ca4da7a/DellEMC_PowerStore_iSCSI/1506 : SUCCESS in 1h 33m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerStore_FC/bea8554/DellEMC_PowerStore_FC/1702 : FAILURE in 4m 57s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerStore_NFS/f83ec66/DellEMC_PowerStore_NFS/1201 : FAILURE in 1h 07m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerStore_NVMeTCP/1787fc8/DellEMC_PowerStore_NVMeTCP/1053 : SUCCESS in 2h 00m 50s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/81/923981/1/check/DellEMC_PowerStore_NVMeFC/f2cd6c5/DellEMC_PowerStore_NVMeFC/17 : SUCCESS in 1h 01m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b4f40a0e033b3a7f675936e4d222ab1332e3be2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-11 23:26:39.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-81-923981-1/FC/ : SUCCESS in 2h 44m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-81-923981-1/iSCSI/ : FAILURE in 5h 33m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"435b451e7bf71d1155df90c18f90f4502a8ea8fb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-12 14:06:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/923981/1/20240712190505/NEC-V-FC : SUCCESS in 3h 57m 06s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/923981/1/20240712190505/NEC-V-ISCSI : SUCCESS in 4h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"630d1e039d096192cb95893a0e4c4f15ce81235c","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-07-15 08:42:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65626e135c9a24c0b59738ae29c4aa0ded2eaacf","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-07-31 19:45:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a874a4400344138b206f26cbb011d486d5150391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:46:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b77bee076050cc6527ed18fc14c2bf959043ac6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:21:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3366b952a40643b6bf8120ae09f1f8a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfd9ae07cd748a2872224ccbc4383d6 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6113815a70544c15962048a720f9a432 : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a180c39f3b9b4294a4ec41d2534cac09 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7965a8b2ffeb4a0aadf390f2c262aa21 : SUCCESS in 12m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46c43d7be53b45cdb5b888ee6a38521d : SUCCESS in 34m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6d505e10b51424f914408beb221b35d : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a3f6763a1c1b9f94341658f8802bc5a289c35e5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:21:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c859c93b1051ae2aa1a680d2fba75f493d10cc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:22:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471e5eae7e2243c1974e7080c9cfa780\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c699410eea4879aa2eb88e8b6187bd : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5b24928c4fa240d9634a49ec1806e335cdf4ff0","revisions":{"d5b24928c4fa240d9634a49ec1806e335cdf4ff0":{"kind":"REWORK","_number":1,"created":"2024-07-11 17:52:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/923981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/923981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/923981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/923981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/923981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/923981/1"}}},"commit":{"parents":[{"commit":"3d654c7f79abce32dbc4579b15edaea04447c362","subject":"Merge \"Remove cyclical import in dell_emc powerstore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d654c7f79abce32dbc4579b15edaea04447c362"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-11 17:48:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-11 17:50:34.000000000","tz":-240},"subject":"Tests: Mock sleep in fungible unit tests","message":"Tests: Mock sleep in fungible unit tests\n\nThis saves a couple of seconds in unit test runs.\n\nChange-Id: I41d706728b2f313378616d005d9069361886f995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5b24928c4fa240d9634a49ec1806e335cdf4ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5b24928c4fa240d9634a49ec1806e335cdf4ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
