)]}'
{"id":"openstack%2Fcinder~924125","triplet_id":"openstack%2Fcinder~master~I51412b12730f07ee8834ffdd2b165f12b54d0e7d","project":"openstack/cinder","branch":"master","topic":"mda","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2025-10-17 15:23:40.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-10-17 15:03:04.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-17 15:23:40.000000000","reason":"Change was submitted"},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-10-17 15:23:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51412b12730f07ee8834ffdd2b165f12b54d0e7d","subject":"[api-docs] Add missing parameters to list attachments","status":"MERGED","created":"2024-07-15 07:01:26.000000000","updated":"2025-10-17 15:24:42.000000000","submitted":"2025-10-17 15:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924125-mda","meta_rev_id":"df9e4c44cff8165c15c39855d014a71e4de8c92e","_number":924125,"virtual_id_number":924125,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 15:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-17 15:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-03-12 05:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-15 10:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-10-17 15:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},{"_account_id":37269,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-15 07:06:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-15 07:20:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-15 07:46:20.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-15 08:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-15 08:30:55.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-15 08:52:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-15 09:19:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-15 09:23:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-15 11:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-15 11:19:11.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-15 11:23:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-15 14:12:45.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-15 16:56:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-16 05:29:04.000000000","updated_by":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"reviewer":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"state":"CC"},{"updated":"2024-07-16 12:40:36.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-07-17 01:31:03.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-07-20 19:30:27.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-08-07 21:08:27.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-07 23:28:56.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-20 14:56:08.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-09-25 09:15:24.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-09-25 15:49:01.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-09 04:42:22.000000000","updated_by":{"_account_id":37269,"deleted":true},"reviewer":{"_account_id":37269,"deleted":true},"state":"CC"},{"updated":"2024-10-12 13:56:30.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-07 23:27:49.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-23 13:21:19.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-03-11 12:10:56.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2025-03-15 10:49:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"440b9ff46a01511f8525e14af0e1abe478a7aad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-15 07:01:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9a6c4b32fbc12832fa9f2212c9be2aff8127236","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-15 07:06:17.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7eabb25bce6b3087780e159c401805f4e65b9b18","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-15 07:20:23.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/eba6e4c516e84565b65f0ad1eedee292\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6894d86407f74e4aa1eaef677e999334 : POST_FAILURE in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"093ca746f659a6dca7b99ea12ede010c883dd3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-15 07:32:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f665baa944859812fdfac50d3e003beaa6f2ea1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-15 07:44:52.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/02f432feb29d43d086ec68f19571847d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6cc759329c3f4c50bf546662e42c794b : POST_FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":2},{"id":"bdcf455bcd771de8ffdb8e99443c90d12fc0c781","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-15 07:46:20.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ec2b9cf21aeb0b049c9c01a0dcb348d984308efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 08:25:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3ea8ff267b4f1c8087accd2b2ec346\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/269908334c744b69ace74b2d4a3cddb8 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1635ac56532244d39c022b8e0923fbb8 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2769bcec56ac4185b534ca56160746c2 : SUCCESS in 11m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9ef3adc152a04c9f94a7ba0c29eb1b26 : SUCCESS in 4m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4335c95d33d64f65ba68cbc1d64fd9d6 : SUCCESS in 5m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a55645667ee402eb509b6269d2bb6e5 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe4ded51d3c44557b306b59f6785ff19 : FAILURE in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7e1cb754cf1261777cfda3255b29cdf0e106df6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-15 08:30:55.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_07_2024_08_25_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"453ee0f83c1a43760dc4fd4fdd39614d52492072","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-15 08:52:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/28edf33/ : RETRY_LIMIT in 8m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/2/upstream-check/cinder-tempest-plugin-ontap-nfs/1437386/ : RETRY_LIMIT in 8m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"50a1cb52c014908f5b7a42e90c53327104c8203a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-15 09:19:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_Unity_FC/d040f3d/DellEMC_Unity_FC/2223 : SUCCESS in 1h 35m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_Unity_iSCSI/9bc8fbe/DellEMC_Unity_iSCSI/2232 : SUCCESS in 1h 47m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3f6cd0244a9cea73e01147737a0096bfb2d8690","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-15 09:23:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerFlex_3.6/432f485/DellEMC_PowerFlex_3.6/1708 : SUCCESS in 1h 46m 29s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerFlex_4.5/1ccd45b/DellEMC_PowerFlex_4.5/589 : SUCCESS in 1h 50m 36s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/28f1917/DellEMC_PowerFlex_4.x_NVMeTCP/544 : FAILURE in 13m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5d2daddb8ac8d7d6c46abb3a367f2470c53789b9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-15 11:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/2/fc : SUCCESS in 0h 45m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/2/iscsi : SUCCESS in 0h 56m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f357c70c8111697b0768d5c11dda1336e898b664","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-15 11:19:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerStore_iSCSI/6e84011/DellEMC_PowerStore_iSCSI/1512 : SUCCESS in 1h 23m 11s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerStore_FC/9cd52d7/DellEMC_PowerStore_FC/1709 : SUCCESS in 1h 11m 33s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerStore_NFS/18d527d/DellEMC_PowerStore_NFS/1207 : FAILURE in 1h 17m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerStore_NVMeTCP/544a41c/DellEMC_PowerStore_NVMeTCP/1059 : SUCCESS in 1h 53m 13s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/2/check/DellEMC_PowerStore_NVMeFC/6e1ce6e/DellEMC_PowerStore_NVMeFC/25 : FAILURE in 2h 53m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"366c5c6c7ea2a571040dc552f8420cd0702871c6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-15 11:23:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-2/FC/ : SUCCESS in 3h 50m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-2/iSCSI/ : SUCCESS in 3h 3m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"154a214378d5069dba1d9b9234f24f9e0dc40db5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 14:12:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/728/.","accounts_in_message":[],"_revision_number":2},{"id":"4004e8b5df71b9b85db0318b991311b862683b1b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-15 14:40:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/242/.","accounts_in_message":[],"_revision_number":2},{"id":"cbcc4f04e3de594f2909661a94d803f8b8b46998","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-15 16:56:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/2/20240715201233/NEC-ISCSI : SUCCESS in 5h 19m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/2/20240715201233/NEC-FC : SUCCESS in 5h 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5bc51b757ec5e5a1e27fe5732fedf0d49fd62ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-16 05:28:50.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b3a6ba22ba44a9cb5c2757192ccbfff8309b3405","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-16 05:30:11.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"751f14f855db38e5b7862a46d41d0e1ab31f99c5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-16 05:42:25.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c140504d62f347bcb07cb18da9408c22\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2b5f7ad754a04257a7e519454db38a89 : POST_FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":3},{"id":"805c92d57578a22579e07dbf7e6e05bc21de9d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 05:53:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4426361b08458d9a618b286a68d108\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ccd1bf58bde04d8f86c720072fa5de1a : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8902c696b5d14169a0c119c4aed6eb83 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd30a1943494986af70ce5e0c75f9d5 : SUCCESS in 15m 08s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4fdec1f274b4b06afae965b37bd4d67 : SUCCESS in 4m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53601625675e47b99beb0db69b87961f : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dbbd1afa8ccd45e081d8a10d32e25717 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/463d91f3f5a14ee7b3ee0b9afdd0f619 : FAILURE in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16e136c6434c5dc62d2acb57f036a7fcac9946fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-16 05:59:05.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_07_2024_05_53_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"0e55a54d125936433d942a6f7d5553d006a4a745","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-16 06:08:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/d059553/ : RETRY_LIMIT in 8m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/3/upstream-check/cinder-tempest-plugin-ontap-nfs/a5b5115/ : RETRY_LIMIT in 8m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7cba0c8c2b2c9d3abf2a3374320131b9c11cc88f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-16 06:23:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/252/.","accounts_in_message":[],"_revision_number":3},{"id":"0e9c66bdaaf9df87153d37adde2a97f3313bec93","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-16 06:50:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/737/.","accounts_in_message":[],"_revision_number":3},{"id":"bb75b5908dfffc2902d1b5485528aba6315ffeaa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-16 08:31:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_Unity_FC/3998777/DellEMC_Unity_FC/2237 : SUCCESS in 1h 28m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_Unity_iSCSI/ad1acb7/DellEMC_Unity_iSCSI/2246 : SUCCESS in 1h 46m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c538023dd1b8a806921138f907690dbb0d68ca0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-16 09:12:54.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerFlex_3.6/21abe35/DellEMC_PowerFlex_3.6/1722 : SUCCESS in 1h 51m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerFlex_4.5/f62ccc8/DellEMC_PowerFlex_4.5/603 : SUCCESS in 2h 00m 43s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/54d8c5f/DellEMC_PowerFlex_4.x_NVMeTCP/558 : FAILURE in 12m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"025e997be51f5e278f0289b30b14cb7ceb2b9ae9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-16 09:38:54.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerStore_iSCSI/fc56750/DellEMC_PowerStore_iSCSI/1515 : SUCCESS in 1h 30m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerStore_FC/3f0cc32/DellEMC_PowerStore_FC/1712 : SUCCESS in 1h 00m 21s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerStore_NFS/55a718e/DellEMC_PowerStore_NFS/1210 : FAILURE in 1h 16m 00s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerStore_NVMeTCP/2047ef3/DellEMC_PowerStore_NVMeTCP/1062 : SUCCESS in 1h 48m 36s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/3/check/DellEMC_PowerStore_NVMeFC/0ee0c32/DellEMC_PowerStore_NVMeFC/28 : FAILURE in 2h 52m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"84b013c0c77c684df21820117f23da1153efbc0d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-16 12:40:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3027f936b7e1e918ee7c9308721b3ed32bb16552","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-17 01:31:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/3/20240717064714/NEC-V-FC : SUCCESS in 3h 43m 19s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/3/20240717064714/NEC-V-ISCSI : SUCCESS in 3h 43m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4a7056f8e03e42f78adcf944efe9dddd832a9554","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-17 06:13:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/3/20240717065831/NEC-FC : SUCCESS in 5h 42m 33s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/3/20240717065831/NEC-ISCSI : SUCCESS in 8h 15m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e18fb1b2bbe5f45421207f041e36c22c08e9b6fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-08-07 21:08:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"13203c89a301f4bc14c451aa102d741409d413e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-07 21:08:27.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/248/","accounts_in_message":[],"_revision_number":4},{"id":"2aa08ca9795eaa96e9e390b935d12540a8dee2bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-07 21:08:29.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/248/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"9afcb706be60e54fb7485278c835f47fe274d733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 21:29:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6c4f788182048959bcf48aeb697f840\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff2c9fe8781b464d9a7ef48282da27e6 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81f9e73394143bcb1e02ed540f63c52 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad370b77b044eb7839557d918117003 : SUCCESS in 13m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4b22f2524854ae5891935c0785940e7 : SUCCESS in 4m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fa53516edf64db5b673b38d79ea9a42 : SUCCESS in 5m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f1568266568d428fb4891e5733a6135a : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67a3278425924957888d87e25d333d62 : FAILURE in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac552b7e696890beb9c86164d9b7cd46f82940b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-07 21:35:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_08_2024_21_29_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"4bdb09b9c99eea40e3e23775ccf30ed286078803","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-07 21:38:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/a35f926/ : RETRY_LIMIT in 3m 39s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/4/upstream-check/cinder-tempest-plugin-ontap-nfs/3618e9d/ : RETRY_LIMIT in 7m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"98a63cfbaf633e21e1a4ca9e060cddc98cfdb680","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-07 23:28:56.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/25/924125/4/check/fujitsu-eternus-dx-iscsi/fb2802f : SUCCESS in 2h 19m 14s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/25/924125/4/check/fujitsu-eternus-dx-fc/6836b15 : SUCCESS in 2h 19m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d03aeee3dda314b1dc7f2a21e22e929907779d7f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-08 12:14:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerStore_iSCSI/529d322/DellEMC_PowerStore_iSCSI/1595 : SUCCESS in 1h 20m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerStore_FC/c872570/DellEMC_PowerStore_FC/1790 : FAILURE in 2h 52m 49s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerStore_NFS/094f312/DellEMC_PowerStore_NFS/1288 : FAILURE in 1h 17m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerStore_NVMeTCP/cc5ae4a/DellEMC_PowerStore_NVMeTCP/1147 : SUCCESS in 2h 07m 38s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerStore_NVMeFC/4b83c18/DellEMC_PowerStore_NVMeFC/111 : SUCCESS in 1h 12m 22s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"b11595cf2c654d057a8d52c1e4d2149450e908bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-08 16:37:03.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_Unity_FC/1dce6af/DellEMC_Unity_FC/2387 : SUCCESS in 1h 47m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_Unity_iSCSI/f9440c9/DellEMC_Unity_iSCSI/2392 : SUCCESS in 1h 59m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"7c1eea3844f27aca4737f8c5874f0ba706a26f27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-08 18:41:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerFlex_3.6/c620e9b/DellEMC_PowerFlex_3.6/1873 : SUCCESS in 1h 54m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerFlex_4.5/bd50aac/DellEMC_PowerFlex_4.5/752 : SUCCESS in 2h 13m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/7f1fb18/DellEMC_PowerFlex_4.x_NVMeTCP/718 : FAILURE in 12m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b9deb20f2b916f82f58ced8ee218c7d0e7c91c4f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-10 06:36:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/4/20240810071722/NEC-ISCSI : SUCCESS in 5h 16m 54s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/924125/4/20240810071722/NEC-FC : SUCCESS in 8h 19m 01s","accounts_in_message":[],"_revision_number":4},{"id":"78b9111e2b7cb0a370f2eb32da7cb326ceb04855","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-13 14:39:57.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/871/.","accounts_in_message":[],"_revision_number":4},{"id":"a04cdd75c9f7a97454e5a717c384ed23edbf2a92","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-17 18:34:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/4/20240817233319/NEC-V-ISCSI : SUCCESS in 3h 48m 05s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/4/20240817233319/NEC-V-FC : SUCCESS in 4h 01m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d7307ce16bfce94ece2f97b4acc8e8286fd7ff9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-09-05 19:58:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ee8da765c578ad4287ac0bbec1d0370c23b1ea1a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-05 19:58:32.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/489/","accounts_in_message":[],"_revision_number":5},{"id":"eede8951474151b947a9e8bea1647606b4f2139c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-05 19:58:37.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"92b5cda2c5eb0e806453f022110b736e44c1726a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 20:24:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd52300f72b46f299020f062b38d7a9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/edfb938f879441a98f42780f80aec5f8 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95eb1f27960439c9cfdf9c91bfc59ec : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b453f2c143f474eb07b81b67cab8b8d : SUCCESS in 11m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/97c3032082c24c9f929d3928504a44ee : SUCCESS in 4m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96cd8e973d1b4d95a4f44282084e392c : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/993b75ae2ef9452593c9513cc99a1494 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/83ddb489a13345c1abc8c9085548b3a2 : FAILURE in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0a4e4d968952c1474b920405c131f48b3aea4385","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-05 20:30:33.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_09_2024_20_25_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"53e16a470b9c334b45b10ea4f524bb30ea7f415b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-05 20:51:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/492/.","accounts_in_message":[],"_revision_number":5},{"id":"5a32542b1ad462cc61fd93e967fc34e20443ec16","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-05 21:24:17.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1080/.","accounts_in_message":[],"_revision_number":5},{"id":"d20afc6581ff95371f4a922087946d9894412ef6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-05 21:26:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/333959a/ : SUCCESS in 50m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/5/upstream-check/cinder-tempest-plugin-ontap-nfs/d9d23f8/ : SUCCESS in 48m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b4d42bad4fe9fca4f965a4ee0d27cf156a346c67","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-09-05 21:38:35.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/25/924125/5/check/fujitsu-eternus-dx-iscsi/cfaf120 : FAILURE in 1h 35m 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/25/924125/5/check/fujitsu-eternus-dx-fc/b35c655 : FAILURE in 1h 39m 31s","accounts_in_message":[],"_revision_number":5},{"id":"b1061446bcda69cd8c57d24494b437da92d178e6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-05 22:14:56.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerStore_iSCSI/d3fbecd/DellEMC_PowerStore_iSCSI/1713 : SUCCESS in 1h 21m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerStore_FC/6656e85/DellEMC_PowerStore_FC/1911 : SUCCESS in 1h 11m 28s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerStore_NFS/dbdb3fb/DellEMC_PowerStore_NFS/1406 : SUCCESS in 1h 11m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerStore_NVMeTCP/2f49913/DellEMC_PowerStore_NVMeTCP/1266 : SUCCESS in 1h 49m 40s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerStore_NVMeFC/0d65ffa/DellEMC_PowerStore_NVMeFC/230 : SUCCESS in 1h 15m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"2cd6809b8bd67f1d8378a3dfcdd029cbbf8f66ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-06 03:25:26.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_Unity_FC/c282a29/DellEMC_Unity_FC/2563 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_Unity_iSCSI/dac084a/DellEMC_Unity_iSCSI/2568 : FAILURE in 2h 35m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c39a18a45807aadfb64d73d76687375c8634b1dd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-07 16:28:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/5/20240907212655/NEC-V-ISCSI : SUCCESS in 3h 46m 06s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/5/20240907212655/NEC-V-FC : SUCCESS in 4h 01m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c846ad3627b74d86e460edbd45ea50f137919fb4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-08 11:07:19.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerFlex_3.6/c0ce921/DellEMC_PowerFlex_3.6/2058 : SUCCESS in 2h 18m 14s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerFlex_4.x/5af79b9/DellEMC_PowerFlex_4.x/941 : SUCCESS in 1h 59m 04s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/41358a7/DellEMC_PowerFlex_4.x_NVMeTCP/896 : FAILURE in 5h 54m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"bef97a7df711f2b1c491c16d66c536873d82ca4f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-08 12:54:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/5/20240908160637/NEC-ISCSI : SUCCESS in 5h 21m 13s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/5/20240908160637/NEC-FC : SUCCESS in 5h 47m 31s","accounts_in_message":[],"_revision_number":5},{"id":"daf7c7d3d03eaa52f60a569c2117c33224bf000e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-20 14:56:08.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"d64af1d01248cfc64b946756d9489a1d9f10b20a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-25 09:15:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/924125-5/Seagate-iSCSI/ : SUCCESS in 2:20:56\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/924125-5/Seagate-FC/ : FAILURE in 6:17:55","accounts_in_message":[],"_revision_number":5},{"id":"b7b327ff9f914d69cfd36957800e931d9b1fbb2c","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-25 15:49:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/924125-5/Lenovo-iSCSI/ : SUCCESS in 2:22:20\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/924125-5/Lenovo-FC/ : FAILURE in 6:17:44","accounts_in_message":[],"_revision_number":5},{"id":"f4d8bead9f5c6ba92529a2cff9c19a3c734ae6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-09 04:42:09.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d685c0aeb8dcfe2beeeabe8d177fee029f17b690","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 04:42:19.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/718/","accounts_in_message":[],"_revision_number":6},{"id":"d8642264e251eae9739ce85e2645f6980fd575f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-10-09 04:42:22.000000000","message":"Patch Set 6:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/43/","accounts_in_message":[],"_revision_number":6},{"id":"7bc68b3458e49a4f2cba62ced1a9bd3224465e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 05:05:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d50ba6b4d44a4572be1b33b71d2681f2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e3ed6c61c1d4b15b41ada57e3a9224c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47a35d3095e45d8a1b7b44364cce7d4 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e82c9635754a26b544272c5cf3d35a : SUCCESS in 11m 22s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9641e17323b047dbb8d4b00d9d516b7a : SUCCESS in 4m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/408776519a9b481ca6a90842ba8e3ff0 : SUCCESS in 4m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c620debd505c4108a2034e8a67aca6ca : SUCCESS in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f58d77c1714408cb6c3766d8bcf8ede : FAILURE in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"30dc32b79ba69f9673f1120762e74a74d2fd56a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-09 05:12:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_10_2024_05_06_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"bdb2b914f4eaa07f01e008aadbda2c6c3a41c272","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-09 05:22:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/4be2b60/ : RETRY_LIMIT in 8m 11s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"785356e24d9bbd95516a1ad59adcc073bcbb48ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 05:42:05.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"f5e36ac4787cd24d0cdf716a3aac923d34a0cec6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-09 07:54:37.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerStore_iSCSI/86633b0/DellEMC_PowerStore_iSCSI/1808 : SUCCESS in 1h 27m 01s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerStore_FC/b9d1970/DellEMC_PowerStore_FC/2006 : FAILURE in 2h 45m 23s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerStore_NFS/9d83f02/DellEMC_PowerStore_NFS/1501 : SUCCESS in 1h 10m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerStore_NVMeTCP/36f2360/DellEMC_PowerStore_NVMeTCP/1360 : FAILURE in 2h 48m 23s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerStore_NVMeFC/2bac0cc/DellEMC_PowerStore_NVMeFC/325 : SUCCESS in 1h 17m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"d4a4e6c6e334191957855e108c7ff30c8dc6fff9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-11 04:59:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/6/20241011100154/NEC-V-ISCSI : SUCCESS in 3h 47m 36s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/6/20241011100154/NEC-V-FC : SUCCESS in 3h 57m 32s","accounts_in_message":[],"_revision_number":6},{"id":"6288628e368ee436207f7b92a45ea967328779ca","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-11 16:55:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/6/20241011172134/NEC-ISCSI : SUCCESS in 5h 12m 19s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/6/20241011172134/NEC-FC : SUCCESS in 8h 33m 28s","accounts_in_message":[],"_revision_number":6},{"id":"0c62b6cb648963d4a5568640eac900376d960c27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-11 18:53:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerFlex_3.6/10ae6d4/DellEMC_PowerFlex_3.6/2223 : SUCCESS in 2h 19m 33s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerFlex_4.x/730fbd4/DellEMC_PowerFlex_4.x/1105 : SUCCESS in 1h 56m 41s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/22faaf2/DellEMC_PowerFlex_4.x_NVMeTCP/1033 : FAILURE in 5h 53m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"95198477185beba0fd33dfc25203e5a39b5d3547","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-12 00:00:10.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/6/fc : SUCCESS in 0h 49m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/6/iscsi : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":6},{"id":"9e8b738fb5fe69440f72e595f84051a5721fc9ba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-12 13:48:50.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_Unity_FC/8c1bbe1/DellEMC_Unity_FC/2724 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/6/check/DellEMC_Unity_iSCSI/5464c83/DellEMC_Unity_iSCSI/2723 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"3b000fcbf74b16e38b42f8eb8038f4b9e2d1fd67","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 13:56:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7c1d131c9d9decc04a71b59bd1e9a679eecbe45f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-13 18:15:11.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"903039390de5c32e2c23d3c1dd26c2169d562752","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-13 18:15:21.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/732/","accounts_in_message":[],"_revision_number":7},{"id":"47ac46880d274ddcdc370e68a984f18b993cd60e","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-13 18:15:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d19bc7bf6d88f9e07c45fb2f8232f5e0dbca5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 18:36:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbfebf07a7b245f98df44ccfb0d30c7d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/95bfa792946e40e082c8a4f429ee14df : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ea2dcfa67841e996e7269b427e9950 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f631407d9bf64ce795bd1cea2c5c4c30 : SUCCESS in 12m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b072ffdbeaf14a848a4e91612eae6911 : SUCCESS in 4m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/762600d0b6004c74a4d217def97e7007 : SUCCESS in 5m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f28278a8720e4815b36cca8ff7284310 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/35d4cc41483a49079db708c094782172 : FAILURE in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a24ea65e400b90528177be4c07c80dbb95fba55c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-13 18:42:38.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_10_2024_18_36_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"30424c896d4a847d44b09ccd11eb8971cc152ea6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-13 18:53:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/b2c4ebc/ : RETRY_LIMIT in 5m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/924125/7/upstream-check/cinder-tempest-plugin-ontap-nfs/5835d4c/ : RETRY_LIMIT in 5m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"5ad466e2ba54b667065cb8eae1353eccda33289e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-13 19:16:20.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"fceaa679f63e64a91b86f54e0240efbe552bdad1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-13 21:00:10.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/7/fc : SUCCESS in 0h 47m 46s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/924125/7/iscsi : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":7},{"id":"967961a3ffb85e63e0fd4c17bd51a941098ac92d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-13 21:22:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerStore_iSCSI/0989577/DellEMC_PowerStore_iSCSI/1816 : SUCCESS in 1h 40m 18s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerStore_FC/59f77f6/DellEMC_PowerStore_FC/2014 : FAILURE in 2h 45m 56s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerStore_NFS/ef2c4ae/DellEMC_PowerStore_NFS/1509 : SUCCESS in 1h 12m 16s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerStore_NVMeTCP/e2e6ee8/DellEMC_PowerStore_NVMeTCP/1368 : SUCCESS in 2h 05m 16s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerStore_NVMeFC/24e0228/DellEMC_PowerStore_NVMeFC/333 : SUCCESS in 1h 19m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"1dcd1d908cb04ee099cd4d215b6b58d736e82721","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-14 06:46:18.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerFlex_3.6/1a25167/DellEMC_PowerFlex_3.6/2234 : SUCCESS in 2h 20m 09s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerFlex_4.x/27b94bc/DellEMC_PowerFlex_4.x/1116 : SUCCESS in 2h 15m 24s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/343b37e/DellEMC_PowerFlex_4.x_NVMeTCP/1042 : FAILURE in 5h 54m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"fa8227ad6519c4a23971b76305919980e96d5bf1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-14 22:49:49.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_Unity_FC/4b6a5ec/DellEMC_Unity_FC/2735 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/7/check/DellEMC_Unity_iSCSI/5050fff/DellEMC_Unity_iSCSI/2732 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4462f6719be1f2e244ddc252c54d9f2e92e948e3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-15 12:19:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/7/20241015173201/NEC-V-ISCSI : SUCCESS in 3h 41m 41s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/7/20241015173201/NEC-V-FC : SUCCESS in 3h 47m 53s","accounts_in_message":[],"_revision_number":7},{"id":"392b3d8274ca6fe76b5df4e406b1c22d661c6f22","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-15 22:57:20.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bc9eada0098fb142a6ae4e6542bc71578576c833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-03 10:40:28.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6ceaea62eff591c5f466db0ff2a3b947dc42d799","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-03 10:40:38.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1139/","accounts_in_message":[],"_revision_number":8},{"id":"31623f928c233c9ff9623343988629f179733c50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-03 10:41:33.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"e26514c6daa93fcbbbb4a33536125f0a31503383","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-03 10:59:10.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/179/.","accounts_in_message":[],"_revision_number":8},{"id":"a4add4fa810f71631733dd0888921cd6bb8fea7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 11:04:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3971af52ac4d12ba16150b69d0c4e9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d33c661e43e9402495d849bf3ed51d86 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1a0b8ff6df446492bea286f1f771c5 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38045c973fb4b8f84b89f747c0acbfd : SUCCESS in 7m 48s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5745e7fa1d934a689b77b44f3e348ede : SUCCESS in 3m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c426a100693c41afae5d6ea1d7aeb0d2 : SUCCESS in 4m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a0267849c8454509a91cec3cd1c536c7 : SUCCESS in 5m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fcfc81f92080464ca258bff3f909ce02 : FAILURE in 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"168c818a01f6d864b1ba681a0b7184e73cdf7d9b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-03 11:04:46.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/168/.","accounts_in_message":[],"_revision_number":8},{"id":"c6534cb2d2c2dc6a0981d07c04d8fba837553a6b","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-03 11:22:18.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"91a2fa63bb1345a848d202fa13e0700c819d66fe","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2025-01-03 14:46:58.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/8/check/DellEMC_PowerStore_iSCSI/6869dd1/DellEMC_PowerStore_iSCSI/1827 : SUCCESS in 1h 32m 25s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/8/check/DellEMC_PowerStore_FC/630dbdc/DellEMC_PowerStore_FC/2023 : FAILURE in 2h 37m 13s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/8/check/DellEMC_PowerStore_NFS/728c725/DellEMC_PowerStore_NFS/1519 : SUCCESS in 1h 20m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/8/check/DellEMC_PowerStore_NVMeTCP/ba626fd/DellEMC_PowerStore_NVMeTCP/1378 : ABORTED in 3h 00m 03s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/8/check/DellEMC_PowerStore_NVMeFC/beb74c2/DellEMC_PowerStore_NVMeFC/343 : FAILURE in 9m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"a348591621070f3069e34044f30c8d5692e13484","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-03 18:15:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e68c96d050261363e921a758ce433e860657c7b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-07 13:00:12.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2e5db488770e977e1c1671747a17551aa0b08d88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 13:00:20.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1163/","accounts_in_message":[],"_revision_number":9},{"id":"944fc2603c97cf84584e675d5fb0e171f54b905e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 13:19:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1dd48d2081a4b9ebdd33419bb7caa87\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/46cb60b80fb24f4882dfa868649a61e0 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38532a156a541aa8a4ff37ea3baabed : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae58d3658eb4ece9e9211faadb65e4f : SUCCESS in 12m 36s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b49d845d2ca43da82b0e1342ed30b7f : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82bf6831e2f14eb0b49ec1657f83137a : SUCCESS in 4m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5529fe0ff97840309a8de36fb448b2eb : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1d1b1be337b40b6b5b66dca06d6a136 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ec1030fb56d6bfc4c04e64f486091d330710f671","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-07 14:00:08.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"ef78818bff5fe6f05fcacee4064b549ab5a0f062","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 16:23:28.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/197/.","accounts_in_message":[],"_revision_number":9},{"id":"22d9e7af487dfad5e7afbc765c468e6e9cf181d0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-07 18:57:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/186/.","accounts_in_message":[],"_revision_number":9},{"id":"3d028aca9f0003ab89e81ab07d9da8dd50df6d77","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2025-01-07 20:06:56.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/924125/9/check/DellEMC_PowerStore_iSCSI/044f8d9/DellEMC_PowerStore_iSCSI/1835 : SUCCESS in 1h 29m 05s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/924125/9/check/DellEMC_PowerStore_FC/7e0d6a0/DellEMC_PowerStore_FC/2031 : ABORTED in 3h 00m 13s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/924125/9/check/DellEMC_PowerStore_NFS/cd881e3/DellEMC_PowerStore_NFS/1527 : SUCCESS in 1h 22m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/924125/9/check/DellEMC_PowerStore_NVMeTCP/3c4658b/DellEMC_PowerStore_NVMeTCP/None : RETRY_LIMIT\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/25/924125/9/check/DellEMC_PowerStore_NVMeFC/730a317/DellEMC_PowerStore_NVMeFC/351 : FAILURE in 23m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"134836b961ac5fefa8751012d55e6acd97c6ba93","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-07 23:27:49.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/25/924125/9/ : FAILURE in 1h 29m 16s","accounts_in_message":[],"_revision_number":9},{"id":"cff1e99f01721fb0db382deb3ddb0400d9bf3c4b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-12 05:09:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/9/20250112100546/NEC-V-ISCSI : SUCCESS in 3h 43m 59s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/9/20250112100546/NEC-V-FC : SUCCESS in 4h 03m 50s","accounts_in_message":[],"_revision_number":9},{"id":"04b3d356daaf131620aeb5d13e0964ab02afceec","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-13 07:47:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/9/20250113105625/NEC-ISCSI : SUCCESS in 5h 26m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/9/20250113105625/NEC-FC : SUCCESS in 5h 50m 36s","accounts_in_message":[],"_revision_number":9},{"id":"84fa2f7ff59bcf24636871d6b18342190e053ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-01-23 12:54:24.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"01263e6c9debcf057ac1749d28030c8c2f4dd943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:19:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259ecb450c8040db95721ff42900078d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30bdd555703444ca8e0515ae5eb6a0fe : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa5c7fdb59245ff8797083757e70615 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b8c3d6c6c741609078aa9461ca9451 : SUCCESS in 15m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d8217a3f5e9e48b09e1c259305620cfa : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46c7d6f7a5954217ab0715273d7d4114 : SUCCESS in 5m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d0cc2821dd784eec8424f1397341f4e8 : SUCCESS in 4m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a686807dc275414388fae89114d03078 : FAILURE in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"532ece3b2947dee29deb532ddf79e2688ecabb2c","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-23 13:20:35.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"eed1f316c635607bcfc3c9abf57b10930776aefe","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-23 13:20:43.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"7f8d86a6d4ee3a709e6fdbeb76f399bd54e02364","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-23 13:21:19.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"b4075c449df561133bd555bb93b1ff6bb0e7abeb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-23 13:21:25.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"a254c24d6f7ce1a5ea5f5e7e545eefd2824ec9ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-23 13:24:26.000000000","message":"Patch Set 10:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/23_01_2025_13_19_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"103b4aa5b9ade5bee8db9e0a74f0463ae5cfbd6d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-23 14:46:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/25/924125/10/ : FAILURE in 1h 43m 44s","accounts_in_message":[],"_revision_number":10},{"id":"1d75e7e8afdb3773a4441abc6524a6c6f1a5270f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-23 16:55:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/10/20250123215427/NEC-V-ISCSI : SUCCESS in 3h 55m 24s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/10/20250123215427/NEC-V-FC : SUCCESS in 4h 00m 58s","accounts_in_message":[],"_revision_number":10},{"id":"aa95f55ca213d856995539b4a6ae20a2402f742b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-23 17:33:49.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-10/FC/ : SUCCESS in 2h 39m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-10/iSCSI/ : FAILURE in 4h 39m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"a8a6f77f38e04ab6767b9c27fcecea7fbe744ec2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-24 04:37:07.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/10/20250124040909/NEC-ISCSI : FAILURE in 9h 05m 28s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/10/20250124040909/NEC-FC : FAILURE in 9h 27m 55s","accounts_in_message":[],"_revision_number":10},{"id":"dcfc485893c744e21583d21c10415a8375ad20c6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-25 04:12:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/296/.","accounts_in_message":[],"_revision_number":10},{"id":"4cb648607b7d71ce04b691908c7e977d4f22acea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-25 08:42:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/278/.","accounts_in_message":[],"_revision_number":10},{"id":"9020c66fead3f193f92e02545cdefea010674fe6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:51:15.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1254/","accounts_in_message":[],"_revision_number":10},{"id":"127453ef6fb10a4e69b73ebd82515e4167251e06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:51:17.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"8d9e45dc9ea53f01257ced96096832cdd8d0982c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-05 19:53:42.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"8cc6fecf1bbd368a74d361d23da71d78a67338e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-05 19:53:53.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1395/","accounts_in_message":[],"_revision_number":11},{"id":"9ceb522ac7206825abf8dbda368bf9e6426ddfdb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-05 19:53:55.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"dc79bffafeaa3977c2ff7f50ed85a708938b0ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 20:41:15.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52560cfe783e499796b9bb4aac32ce36\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/36aafffb38624a27b8459f2cc9fdb4a9 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877f83e6bd1849a3be52d8bb25bb5b76 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3229ff58a47342098e7997e3ce37e9a3 : SUCCESS in 11m 48s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95bc5ede309d44af942f008e3cf65ce2 : SUCCESS in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43742a9a2f604d2ea158052976bea97d : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/288ddbb4c1d44e0dbb0a442befbb0cbd : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e8ad93d51f2f455c80e72253e4134b95 : FAILURE in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"05ff18470e374fd77ad09fd5f668fa32943bd19f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-03-05 20:42:12.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"5f0f2502f4d36c8c559fb040de630a1d64f504fb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-03-05 20:42:18.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"69514c842f18981828d790228ec5cf9ac8f0a097","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-03-05 20:42:59.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"bef69cb292e279db6d82aa1acb03d60704202de3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-03-05 20:43:07.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"200eb52b4366ff0e9aa7643cba2a083c97664647","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-03-05 20:46:41.000000000","message":"Patch Set 11:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/05_03_2025_20_41_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"fe572b6e8446e508181ddf49f8d987594c67531b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-05 21:51:41.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-11/FC/ : SUCCESS in 1h 57m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-924125-11/iSCSI/ : SUCCESS in 1h 44m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"5cfbb7f80b23f00e0bbec88ebf70a48b9782ffbd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-05 22:42:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/425/.","accounts_in_message":[],"_revision_number":11},{"id":"e2c30f40bb7c0eacee6a980bd72f50ed45b79d25","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-05 23:42:24.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"41c47fbf9516bc0839f3fde5d129caacb3af7f16","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-06 14:49:58.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/11/20250306192826/NEC-V-FC : SUCCESS in 4h 18m 13s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/11/20250306192826/NEC-V-ISCSI : SUCCESS in 4h 21m 29s","accounts_in_message":[],"_revision_number":11},{"id":"568d36c1e5de3e036cfd1d07dcc14e49b713654f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-06 20:59:08.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/11/20250307000403/NEC-ISCSI : SUCCESS in 5h 30m 35s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/11/20250307000403/NEC-FC : SUCCESS in 5h 55m 03s","accounts_in_message":[],"_revision_number":11},{"id":"acee81883cfc8299b1a9d053b1562fcad089476a","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-03-11 12:10:56.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1d6a1e71ea1e582639d5a5c22358ca9c6b793f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-12 05:12:03.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"8b66fc525317be8307a045c0d9841e2c9a5872e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-12 05:12:11.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1458/","accounts_in_message":[],"_revision_number":12},{"id":"0dd3d90293ec0e1ecc1dba410a4251df7b26d817","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-12 05:12:13.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924125/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"6a0252f336bbd7e57824a759ff022e026291ebb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 05:32:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471652887c1f458e87d78dd9e10368c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af0550b7645f42ad9a9227326cffcc82 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5d131222ba41e493bbca187d595a78 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508a412407a34687a40973a7128d71db : SUCCESS in 12m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/32fc6977fcce48d9a84c5957f939f22e : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/766953a4fb084e5ea0800b4929f3742c : SUCCESS in 3m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3f5f674cee55458d876737ac3d84fe78 : SUCCESS in 3m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7f569581802406dbc50baf3c18effa6 : FAILURE in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"51c70f5cd09ee00684548de0abe8d1c39c1eb91b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-12 07:33:54.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"8dfe8143666771430608206430ba005e4a46a2c1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-15 03:58:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/12/20250315044637/NEC-FC : SUCCESS in 5h 56m 40s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/12/20250315044637/NEC-ISCSI : SUCCESS in 8h 11m 40s","accounts_in_message":[],"_revision_number":12},{"id":"3762cb6c1805f09ad47fa324cc78846f9373e5ae","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-15 04:28:50.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/12/20250315090500/NEC-V-ISCSI : SUCCESS in 4h 19m 22s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/25/924125/12/20250315090500/NEC-V-FC : SUCCESS in 4h 23m 47s","accounts_in_message":[],"_revision_number":12},{"id":"74fb3a9cd600f70512fce305ba0940e086a8ce1c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-15 10:49:28.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"236cadaeec396ae3d863f8d3bbdadde9c835c0ad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-17 15:03:04.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a2d29f576f16aecc2b427642b13c82fec43053b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 15:03:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"b761ad6083c6f5bccecac1011555ea61e0da7efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 15:23:39.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639a94d3098544ca8be459a89994c495\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/895306f910a745bfaf8d77627835a0e3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a46b0436cc476cb6f9d0493aaf0b04 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89aefc8296a4612b5d73b77409a27c4 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":12},{"id":"e8f0a623aaa3ec709e1a8277b24fe2a3e5fd1b27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 15:23:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"df9e4c44cff8165c15c39855d014a71e4de8c92e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 15:24:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1856f13b13437885c1aa5e143c3491\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/21b614843d634c8fa5124ae08308d9bb : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd0ade848284b84a855b149dc59e219 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"5f51889139880a4bd5092a9d0f661b67c5b7569e","revisions":{"d217a84904fef15e641b836e8a903c66c14264bc":{"kind":"REWORK","_number":1,"created":"2024-07-15 07:01:26.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/1"}}},"commit":{"parents":[{"commit":"3d654c7f79abce32dbc4579b15edaea04447c362","subject":"Merge \"Remove cyclical import in dell_emc powerstore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d654c7f79abce32dbc4579b15edaea04447c362"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:22.000000000","tz":180},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d217a84904fef15e641b836e8a903c66c14264bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d217a84904fef15e641b836e8a903c66c14264bc"}]},"branch":"refs/heads/master"},"0c7fdf777374839aed16c0d3084ca5b379f3f0a7":{"kind":"REWORK","_number":2,"created":"2024-07-15 07:32:25.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/2"}}},"commit":{"parents":[{"commit":"3d654c7f79abce32dbc4579b15edaea04447c362","subject":"Merge \"Remove cyclical import in dell_emc powerstore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d654c7f79abce32dbc4579b15edaea04447c362"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:32:10.000000000","tz":180},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c7fdf777374839aed16c0d3084ca5b379f3f0a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c7fdf777374839aed16c0d3084ca5b379f3f0a7"}]},"branch":"refs/heads/master"},"18856517ee425ced8fc6199b85f0b4137fd5e946":{"kind":"REWORK","_number":3,"created":"2024-07-16 05:28:50.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/3"}}},"commit":{"parents":[{"commit":"3d654c7f79abce32dbc4579b15edaea04447c362","subject":"Merge \"Remove cyclical import in dell_emc powerstore driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d654c7f79abce32dbc4579b15edaea04447c362"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-16 05:28:50.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18856517ee425ced8fc6199b85f0b4137fd5e946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18856517ee425ced8fc6199b85f0b4137fd5e946"}]},"branch":"refs/heads/master"},"f691261d9d27f6d772895d02934bd5509239297d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-07 21:08:16.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/4"}}},"commit":{"parents":[{"commit":"30cdc86784bc211c6ac3071b20460436699dd217","subject":"Merge \"More thorough mock on a backup message test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30cdc86784bc211c6ac3071b20460436699dd217"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-08-07 21:08:16.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f691261d9d27f6d772895d02934bd5509239297d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f691261d9d27f6d772895d02934bd5509239297d"}]},"branch":"refs/heads/master","description":"Rebase"},"2f3d6dea11f8e708f02c2368993c56369cdb9601":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-05 19:58:23.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/5"}}},"commit":{"parents":[{"commit":"25699617a1251d727a83e969fc27a16ce7b91f5b","subject":"Merge \"Dell PowerStore: Added timeout into rest API call.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25699617a1251d727a83e969fc27a16ce7b91f5b"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-09-05 19:58:23.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f3d6dea11f8e708f02c2368993c56369cdb9601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f3d6dea11f8e708f02c2368993c56369cdb9601"}]},"branch":"refs/heads/master","description":"Rebase"},"32f74fdcce43ae8b42ace4d43207cc44a23bd39a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-09 04:42:09.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/6"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 04:42:09.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32f74fdcce43ae8b42ace4d43207cc44a23bd39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32f74fdcce43ae8b42ace4d43207cc44a23bd39a"}]},"branch":"refs/heads/master","description":"Rebase"},"82748f49eb11a34d353bd33f32c2707efb6b6a93":{"kind":"REWORK","_number":7,"created":"2024-10-13 18:15:11.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/7"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-13 18:15:10.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82748f49eb11a34d353bd33f32c2707efb6b6a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82748f49eb11a34d353bd33f32c2707efb6b6a93"}]},"branch":"refs/heads/master"},"1580727aa8ab78d50b11b3a67c964caf2c2442f9":{"kind":"REWORK","_number":8,"created":"2025-01-03 10:40:28.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/8"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-03 10:40:28.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1580727aa8ab78d50b11b3a67c964caf2c2442f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1580727aa8ab78d50b11b3a67c964caf2c2442f9"}]},"branch":"refs/heads/master"},"d5de2f98a4895cd77b9eb028807f3ea0bd982496":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-01-07 13:00:12.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/9"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-07 13:00:12.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5de2f98a4895cd77b9eb028807f3ea0bd982496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5de2f98a4895cd77b9eb028807f3ea0bd982496"}]},"branch":"refs/heads/master","description":"Rebase"},"5ca762f471e008a64207fae93e770db86766cca8":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-01-23 12:54:24.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/10"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-01-23 12:54:24.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ca762f471e008a64207fae93e770db86766cca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ca762f471e008a64207fae93e770db86766cca8"}]},"branch":"refs/heads/master","description":"Rebase"},"50c95c170f9322b78ee2a79f02b4b9bbb9de668f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-03-05 19:53:42.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/11"}}},"commit":{"parents":[{"commit":"64bf9fc0bb2bb39773fdd0feef5be9398243f90a","subject":"Merge \"HPE 3par: Add comment for cloned volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64bf9fc0bb2bb39773fdd0feef5be9398243f90a"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-05 19:53:42.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50c95c170f9322b78ee2a79f02b4b9bbb9de668f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50c95c170f9322b78ee2a79f02b4b9bbb9de668f"}]},"branch":"refs/heads/master","description":"Rebase"},"5f51889139880a4bd5092a9d0f661b67c5b7569e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-03-12 05:12:03.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/25/924125/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/924125/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/924125/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/924125/12"}}},"commit":{"parents":[{"commit":"79928cd6a9c2e81c6845ba322d156b58071d4a52","subject":"Merge \"Dell PowerMax: Enhanced the retry mechanism to verify the active snapshots.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79928cd6a9c2e81c6845ba322d156b58071d4a52"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-15 07:01:10.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2025-03-12 05:12:03.000000000","tz":0},"subject":"[api-docs] Add missing parameters to list attachments","message":"[api-docs] Add missing parameters to list attachments\n\nIt is possible to filter attachments based on\n * instance_uuid\n * volume_uuid\nAdd those parameters to documentation.\n\nChange-Id: I51412b12730f07ee8834ffdd2b165f12b54d0e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f51889139880a4bd5092a9d0f661b67c5b7569e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f51889139880a4bd5092a9d0f661b67c5b7569e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
