)]}'
{"id":"openstack%2Fcinder~924505","triplet_id":"openstack%2Fcinder~master~If8d91a789dd7619404ca493fda3cd230fdecc651","project":"openstack/cinder","branch":"master","topic":"If8d91a789dd7619404ca493fda3cd230fdecc651","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2024-10-21 04:20:55.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-11-15 18:44:09.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-11-15 09:18:07.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-11-15 18:44:09.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-11-15 16:21:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7198\u003e replied on the change","reason_account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}},"hashtags":[],"change_id":"If8d91a789dd7619404ca493fda3cd230fdecc651","subject":"Nimble: Report max_oversubscription_ratio via backend capabilities","status":"MERGED","created":"2024-07-19 12:19:01.000000000","updated":"2025-01-13 17:06:30.000000000","submitted":"2024-11-15 18:44:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924505-If8d91a789dd7619404ca493fda3cd230fdecc651","meta_rev_id":"cf76f8e4f48d198cf5733c93ba720a68a94cddeb","_number":924505,"virtual_id_number":924505,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 18:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-11-15 12:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-15 16:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-15 16:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-19 12:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 13:13:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-19 13:53:30.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-19 16:09:29.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-19 16:20:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 16:22:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-19 16:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 16:32:26.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-19 16:46:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-19 17:55:39.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 18:58:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-19 19:13:19.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 19:52:46.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-19 19:57:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-19 21:03:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-19 21:17:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-20 08:48:08.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-20 17:10:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-24 13:50:55.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"CC"},{"updated":"2024-07-26 07:46:55.000000000","updated_by":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"reviewer":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"state":"CC"},{"updated":"2024-07-26 08:19:54.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-07-26 13:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-10-01 12:50:48.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-10-15 12:21:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-23 01:31:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-23 01:49:48.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-10-23 04:37:02.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-11-14 18:24:26.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2024-11-14 22:54:08.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-11-15 16:21:27.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"e182666feffd167864276cda52ca9be9e540e0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-19 12:19:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0284a27f1b61e3f207bbe0a7e2fd8007e6e11d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-19 12:20:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"beddb7d6dcef85f9ec7429652b23dae3fbeb6316","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 12:36:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a438a69dcd411686949785a925e7ac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f9a8f918fbf4d4280fac29b9eb3fadb : FAILURE in 9m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38fef99c986b426aacfb6354f1fb37c2 : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"914eda725cc82525ff755603f46e250641590a73","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 13:13:16.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/15fb74db4714474d8746dda45b487451\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c2a31ed76edb46b68d8ec3766f963243 : POST_FAILURE in 28m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6077e12d812ad580a99349bb0368073579149d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-19 13:28:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84954416cf623eb585918cfa1b2bd182a6dbdbb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:46:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd904fd725164613ba0604908ff0cb3b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03750f0ec7b34dc7925d612286a6f40c : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e34b3772ca29425993a1a2ab9b58d4e2 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b6ead49acfa87d3770fd55cc008e3adcdb85febc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-19 13:53:30.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/99/","accounts_in_message":[],"_revision_number":3},{"id":"8062351efedd9920ef9260bbd7b19acdb1de36f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-19 14:16:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/99/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"5cd2b237c042c26f8b210058a56b10fac70ca883","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-19 14:17:22.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cc148993af4e4fffa9ed8154645dfb7a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a02e58213f1b47ab969ad59577bb33e4 : POST_FAILURE in 22m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5cd4ebc72e8a7aa7d8df597ba8615617773d811e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-19 16:09:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924505 : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":3},{"id":"26411c6d1a6f4bed0fc641b35746b7802fb66eba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-19 16:20:14.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924505 : FAILURE in 31m 46s","accounts_in_message":[],"_revision_number":3},{"id":"3da3024ffe4cfac44d599a2fc434078071266b60","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-19 16:22:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 34m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4b65f1f21cb93985aad71546be83e14f8d60178c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 16:26:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd9ca11c81d4ce7a76cb6d2033e621c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee42824587c44bd39c1f5cac3778b665 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86235f316c6c4ea7bd7cb8336fc90c41 : SUCCESS in 11m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6aafa66bf22241faa21de052b0c9b43a : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a20e9edf15574838a8afcf3a543f2745 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efa41bc53414a13baade03ca48a7dd9 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/f784867e0fb7425e88b463e42aa06f98 : SUCCESS in 54m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/457ad5fb8c6b4185a2b58fc6a10a920d : SUCCESS in 1h 34m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ca06cf81e914a2f9592aa180b1d57a4 : SUCCESS in 1h 02m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79da58ead964fda858ed1fad3bc953e : SUCCESS in 18m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fcf73829606848318852043ca5f9b993 : SUCCESS in 17m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/82b1a19ec81d42d4a8029e64dc18483a : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c238674a94e94af682b78394ae904e1f : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24f14128423242b180230e8f3f5e67bc : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b7a8ae7b9e0640f593796de2f55f39c1 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00a35e80ca3c4bd997ae3380cd5dda99 : FAILURE in 39m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36bd45ae8158413aa74848c273b36c9a : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d255a4ea49aa4df790188a6b1e87d4e1 : SUCCESS in 1h 52m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac1684b971bb49efb338310cec4ad493 : FAILURE in 2h 05m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5083c2db948945acb850c0a4adc449de : SUCCESS in 1h 43m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/99cb98335cb24fcda67591f7811361e8 : SUCCESS in 2h 51m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/228f95e6883d4bc798baa803b6effdb4 : SUCCESS in 32m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2ac5f5801b1488dad2bc85e5ba3689b : SUCCESS in 1h 18m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6aabf3631284b86b04119738cf8cabc : SUCCESS in 1h 28m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3910f3d22d1e43ac863e2efbdaf75641 : SUCCESS in 40m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c1c5d511ea94c84bed9ce44c6b55a95 : SUCCESS in 1h 49m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b4daabf0e9cc418c84779483d715cb94 : SUCCESS in 1h 37m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cdd8def258c45da8f7aaca59d2a7c13 : SUCCESS in 1h 40m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da3b565cc27f478fa82df17c9726f786 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":3},{"id":"254a6b4f5b3cd4d432b01690de389daaf9aa3220","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-19 16:32:26.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_07_2024_16_26_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"f4e66bdc60666c07f93b34a88f7a5335d81000b0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-19 16:46:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/924505/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/9e30587/ : RETRY_LIMIT in 3m 21s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/924505/3/upstream-check/cinder-tempest-plugin-ontap-nfs/7cdcef5/ : RETRY_LIMIT in 7m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9b206cd4bf094ad4460c99d13f2f4331c154bed4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-19 17:55:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/924505/3/check/fujitsu-eternus-dx-iscsi/3da5a47 : FAILURE in 2h 09m 30s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/924505/3/check/fujitsu-eternus-dx-fc/f4bc1e9 : FAILURE in 2h 00m 45s","accounts_in_message":[],"_revision_number":3},{"id":"96b4cc220230e85a11083f8b8c9e5bfad800a5e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-19 18:58:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/762/.","accounts_in_message":[],"_revision_number":3},{"id":"12fafda2b1f3a40fc1b6446bf8c9f5e79fbfebda","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-19 19:13:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/21452ad9fbd84df48f1d6196f4ba84a4 : SUCCESS in 1h 11m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cd92a04263364f328135471404a02213 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f9ff0890bf13e28899b91cc1dbfbb42685e9ac02","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-19 19:52:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerStore_iSCSI/81fe308/DellEMC_PowerStore_iSCSI/1528 : SUCCESS in 1h 31m 12s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerStore_FC/add04e4/DellEMC_PowerStore_FC/1725 : SUCCESS in 1h 01m 02s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerStore_NFS/c6b99ab/DellEMC_PowerStore_NFS/1223 : FAILURE in 1h 18m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerStore_NVMeTCP/c59cea1/DellEMC_PowerStore_NVMeTCP/1082 : SUCCESS in 1h 49m 05s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerStore_NVMeFC/39a3355/DellEMC_PowerStore_NVMeFC/44 : FAILURE in 8m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"a8ea60e2011cf73d5d42eba960155fcb35bd6269","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 19:57:26.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-3/FC/ : SUCCESS in 2h 50m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-3/iSCSI/ : FAILURE in 4h 21m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"df2600107e144f54fb612f7669f6a43d1ad6b40a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 20:24:38.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-1/FC/ : FAILURE in 7h 18m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-1/iSCSI/ : UNSTABLE in 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"eb82d88fe86c2245be9c55b78198a53e2fc36b05","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-19 21:03:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerFlex_3.6/d361cba/DellEMC_PowerFlex_3.6/1754 : SUCCESS in 1h 53m 17s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerFlex_4.5/3413d87/DellEMC_PowerFlex_4.5/633 : SUCCESS in 2h 06m 32s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/f3ace5d/DellEMC_PowerFlex_4.x_NVMeTCP/591 : FAILURE in 13m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"86c2f24e14154f62beab9114a3bd33443187fa19","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 21:05:55.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-2/FC/ : FAILURE in 7h 29m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-924505-2/iSCSI/ : SUCCESS in 2h 3m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"49b6efb92271c4ee3420e46e6fd9959272f5ca44","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 21:17:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_Unity_FC/ed896b0/DellEMC_Unity_FC/2266 : SUCCESS in 1h 30m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/3/check/DellEMC_Unity_iSCSI/a692e23/DellEMC_Unity_iSCSI/2273 : SUCCESS in 2h 17m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a11ba16a17a62417f6ca0e391eda596a55e62d7a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-20 08:48:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/3/20240720135730/NEC-V-ISCSI : SUCCESS in 3h 44m 19s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/3/20240720135730/NEC-V-FC : SUCCESS in 3h 50m 34s","accounts_in_message":[],"_revision_number":3},{"id":"e0705d5a455d85e7478189612d116327b631cddd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-20 17:10:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/3/20240720202627/NEC-ISCSI : SUCCESS in 5h 18m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/3/20240720202627/NEC-FC : SUCCESS in 5h 44m 18s","accounts_in_message":[],"_revision_number":3},{"id":"2c006219a74d910543761e6f4d46ae9fe30eb81e","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-21 05:43:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fa9c85ef013f1bc747542393129e5a9b2a93a598","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-21 05:45:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ec5ff047027448c551d75ec4f572c1049adf285a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-21 06:14:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/924505/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/536f529/ : RETRY_LIMIT in 3m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/924505/3/upstream-check/cinder-tempest-plugin-ontap-nfs/20c2413/ : RETRY_LIMIT in 3m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e57368efcd81bf60706c40e077c53652d38d1dd5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-21 06:24:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":3},{"id":"1668dae0fea2e8500346ae1cb04d7ca43e13cef6","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-21 17:10:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22b608b7d57f88bb119cfd110ecfb6ba8347294f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-21 17:49:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 36m 49s","accounts_in_message":[],"_revision_number":3},{"id":"0be173e6dddd6837ed643086fce21888194d648d","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-23 05:06:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4fd82e7f8e5ffbf57c3a260c485625fe17cbd48","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 06:28:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":3},{"id":"05255bc3b11cbdbe87fdac73d3405899e5fea781","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-07-24 13:50:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8c1cb86432ee9b9fae499d5368b9ba63813f7260","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-24 19:08:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86c310edcfcfa6ee0d97bd1a3acf6afe1d736fa3","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-07-25 06:50:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a25ac7e6feca66c84defc9c29dec3e1f1668e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-26 07:46:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a83059f028c65d043055114853a0b74a9c855af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-26 07:46:54.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"17d770daafba113d71518f6842871e734a69cd94","author":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"date":"2024-07-26 07:46:55.000000000","message":"Patch Set 4:\n\nBuild SUCCESS.\n\nBuild URL: http://localhost:8080/job/cinder/32/\nDuration: 1.1 sec and counting","accounts_in_message":[],"_revision_number":4},{"id":"7c4232439bb1da70bae0e30138466f30fdba8796","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:46:56.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/155/","accounts_in_message":[],"_revision_number":4},{"id":"09675f9db4d7f26065b66fd87f1a0e23ec4cd05d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:46:58.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/155/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"a2025b430dd317ee0b0ba8cbc6ef030c3a8149ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-26 07:47:04.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"098599c04918e89e4332aa2cde36ec6d2a39ffb6","author":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"date":"2024-07-26 07:47:06.000000000","message":"Patch Set 5:\n\nBuild SUCCESS.\n\nBuild URL: http://localhost:8080/job/cinder/33/\nDuration: 1.1 sec and counting","accounts_in_message":[],"_revision_number":5},{"id":"9c483f79f5f294a1305974f75c4d42d0d8c2ffe0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:47:06.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/156/","accounts_in_message":[],"_revision_number":5},{"id":"0767886779bbbd47c1d34ebc72653b5ef81d07d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:47:07.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/156/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"c68b20e21953cfe885e6bbe67d46b151f65b3c48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:47:11.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/157/","accounts_in_message":[],"_revision_number":6},{"id":"1e3f51080ef2e56c97308bbf1e21bb5e58858e43","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:47:13.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/157/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"ccf33d21a5c2d782bc6b9ff82aa02ca119fe4561","author":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"date":"2024-07-26 07:47:21.000000000","message":"Patch Set 6:\n\nBuild SUCCESS.\n\nBuild URL: http://localhost:8080/job/cinder/34/\nDuration: 1.1 sec and counting","accounts_in_message":[],"_revision_number":6},{"id":"29e1834abe95d77824964932e0a9f83eeaafee7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-26 07:47:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a3d09ab67545c5c71641222b2349fcea47544e0b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:48:01.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/159/","accounts_in_message":[],"_revision_number":7},{"id":"5dadc5721c0cb2c333eb5598d3c37ab1e792bf6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 07:48:03.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/159/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"5bb9032f721b6ae6bcc00b62f68477f9f1d21a79","author":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"date":"2024-07-26 07:48:06.000000000","message":"Patch Set 7:\n\nBuild SUCCESS.\n\nBuild URL: http://localhost:8080/job/cinder/36/\nDuration: 1.3 sec and counting","accounts_in_message":[],"_revision_number":7},{"id":"b2c1f89cfe148087d05c15b39c2cadc4a92e5da0","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-07-26 07:48:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b363b423c8eb2f25f55c0043df59b18ed5ae8e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 08:11:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1635d4f5b594eb29ba5ddbbba0b2eaf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d34aa53d039b48bab57eab53a975ea75 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f735dfd1946044abbafa2579929eab78 : SUCCESS in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"96fda6eccce2f26993467af2801c84d9a888e93d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-26 08:13:42.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3b6a70d98cf1484a8bfe0d5734c0ab0b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/62a55c641e3b4e0abdbe8cae822786b9 : POST_FAILURE in 19m 35s","accounts_in_message":[],"_revision_number":7},{"id":"9bfb3a7291f7ec3e2ebc504d5d86c3ad6c251d0b","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-07-26 08:19:54.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"658cc6256cd9a6865fa116308ca109a63768a5e3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-26 09:05:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924505 : FAILURE in 31m 55s","accounts_in_message":[],"_revision_number":7},{"id":"62f2ea00902d511fd9182d51664229fdbb2f68cf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-26 09:32:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 36m 08s","accounts_in_message":[],"_revision_number":7},{"id":"232471424140cc183c053c99371ed0df7eb2add1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-26 09:35:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924505 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":7},{"id":"ca1a7a74e741557025f18aaf57f54e7f650fa3b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-26 10:23:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/82a04ccf0a5d4051bc7b35c5944e5e05 : SUCCESS in 1h 15m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bfe89ed96d9543e2842eb5b329afd23e : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":7},{"id":"156ad3b9e06ebaee717e8e2da52cf86e5bec73c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 10:27:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eaa618e7f88498684854aa1f0e9264a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d48f7273df4bd48de21784d9b2267a : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21ac6071401340f0a2df4f8594329a84 : SUCCESS in 9m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10e291c2619344879abafcc437b31da8 : SUCCESS in 8m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4cfd87fcfd24040b2972fca65d10643 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ea5436c8024885852566ce07153b63 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/2079c3c6ac3a4e75ab334b30bed7cbb2 : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49fcf10cf90f430a9f7f839e83bff15c : SUCCESS in 1h 28m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/531e55889cac426980f5df5a000d68b2 : SUCCESS in 1h 01m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52092e8e05e4b058b2fb759b879b49a : SUCCESS in 14m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f828b3342b2410885b68cbc66a023ad : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/59e941647c0642f3bbe2f7461f05fe82 : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7bfc6454a8cb40439db827f48f231525 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f6c1cd9eff74186bd4ed31900db65cd : SUCCESS in 7m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb1c6a160e914104bac28a42c0c66d04 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d533b0a53754f8a8807ea0486d06bb2 : FAILURE in 41m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19360931fd9e4c2bb6ed624e13af2676 : SUCCESS in 4m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b29a5e1642704ee0bbb54334edaf823a : SUCCESS in 1h 44m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9073a844ee8241bb9c5fddd03f77c2c9 : SUCCESS in 2h 26m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f241eae72a14fc78864a9150d0c590c : FAILURE in 1h 49m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8a6c70dc7a0a4628821938bc007ac387 : FAILURE in 21m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a141fe7bd4024688acb0f268409d85dd : SUCCESS in 36m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a50e28ff28724a2abe1e5ee18f86cc27 : SUCCESS in 1h 14m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/184e7c8c899940af89f3376d78ce18f2 : SUCCESS in 1h 17m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52e44c584a1c4a799028f9f532c961a7 : SUCCESS in 40m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ae06b6ccec604cf3aca1bed1ada8883b : SUCCESS in 1h 29m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1c617cd452bd47a588fb420a6313bd85 : FAILURE in 20m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82d5c1b5b76c4c3798ad39e54d030142 : SUCCESS in 1h 34m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e530c7383294e7b9b1edae37fa1f5ac : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":7},{"id":"2aa2d1307efb43dd726343515d5554721de6fe04","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-26 11:49:02.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/7/check/DellEMC_Unity_FC/570f3f6/DellEMC_Unity_FC/2307 : SUCCESS in 1h 38m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/7/check/DellEMC_Unity_iSCSI/f0e1d31/DellEMC_Unity_iSCSI/2312 : SUCCESS in 1h 53m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"9bdb10e499fd231398d735b4b0e8dcc848ab9cfc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-26 12:50:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/801/.","accounts_in_message":[],"_revision_number":7},{"id":"59773ae15b50ebb60956d60a262040e2876fb613","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-26 13:00:08.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/7/fc : SUCCESS in 0h 52m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/7/iscsi : SUCCESS in 0h 58m 13s","accounts_in_message":[],"_revision_number":7},{"id":"6565c521c0d7d72b49520acc4714d6e28706035c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-26 13:50:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/7/check/DellEMC_PowerFlex_3.6/de5350e/DellEMC_PowerFlex_3.6/1793 : FAILURE in 2h 53m 45s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/05/924505/7/check/DellEMC_PowerFlex_4.5/08038d6/DellEMC_PowerFlex_4.5/672 : SUCCESS in 2h 13m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/0a0d0c7/DellEMC_PowerFlex_4.x_NVMeTCP/632 : FAILURE in 14m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"109b97781cad6024d11105f86f497330196d13ef","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-26 19:56:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/7/20240727011228/NEC-V-FC : SUCCESS in 3h 38m 16s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/7/20240727011228/NEC-V-ISCSI : SUCCESS in 3h 44m 03s","accounts_in_message":[],"_revision_number":7},{"id":"ac0f501e496dffe84c7061bd3797874a2bd42caa","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-27 16:24:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/7/20240727193735/NEC-ISCSI : SUCCESS in 5h 23m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/7/20240727193735/NEC-FC : SUCCESS in 5h 46m 51s","accounts_in_message":[],"_revision_number":7},{"id":"06cc36f220fd9ad6f5589a4f0dc0a1ccf477f021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-08-05 09:25:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a290bbf27e2662deb5b532b48d4ecaa11053698f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 09:25:50.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/217/","accounts_in_message":[],"_revision_number":8},{"id":"094ed71d9f03674c2845fc5502c249012198c06d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 09:25:52.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/217/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"ddaee3b3cf5c72a0492a78a6c324383fc286d403","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-05 09:30:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"fcf2929d5c11f5ec0ef671536d31949009214de7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-05 09:49:25.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6f99ac052fa941428b412048aa25bafc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3d92120e7d564720818a3305a0f4ae65 : POST_FAILURE in 20m 38s","accounts_in_message":[],"_revision_number":8},{"id":"2f493362071ef59de2bf40228ea03a0cf52b11fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 09:57:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe9029e93a74e11b034277bf15ff3bd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/febd26a4bdb842bfaf0947475f29fb1b : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/053fedf22be4454c9d8525216525ec7c : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f30633fd71c4b8471f52e50283f190c96909afa5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-05 10:05:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 34m 03s","accounts_in_message":[],"_revision_number":8},{"id":"cb8d5ba5680bb246d05b25d96b58bbbdd6079776","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-05 10:07:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924505 : FAILURE in 32m 14s","accounts_in_message":[],"_revision_number":8},{"id":"04a1024280a2ace97b5fda09e519b0cbeaf0a44b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-05 11:32:08.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_3.6/b3b5721/DellEMC_PowerFlex_3.6/1841 : SUCCESS in 2h 04m 17s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_4.5/cc36d2a/DellEMC_PowerFlex_4.5/720 : FAILURE in 2h 01m 57s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/17f59f2/DellEMC_PowerFlex_4.x_NVMeTCP/684 : FAILURE in 13m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"4a38759a0e84b837faaa595df25656015a747f6c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-05 11:38:45.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_Unity_FC/d37305b/DellEMC_Unity_FC/2355 : SUCCESS in 1h 45m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_Unity_iSCSI/9f1e761/DellEMC_Unity_iSCSI/2360 : SUCCESS in 2h 04m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"e6e4ce170b151997c71e0d149a88538db34647d2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-05 11:55:28.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/924505/8/check/fujitsu-eternus-dx-iscsi/e66e175 : SUCCESS in 2h 27m 06s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/924505/8/check/fujitsu-eternus-dx-fc/7f4c82c : SUCCESS in 2h 25m 43s","accounts_in_message":[],"_revision_number":8},{"id":"65068e0eb8e51eabd380ca55c9812b5c3997c8d5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-05 12:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/8/fc : SUCCESS in 0h 35m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/8/iscsi : SUCCESS in 0h 51m 21s","accounts_in_message":[],"_revision_number":8},{"id":"be9555c2ee66df69a4dcc9d4b45b984c3e823573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:07:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a642a5025f4fdbb1b8207d9a063302\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d735444a324f4c259f1086f23a9684d0 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8db5e7baa669438c818895c720ff3ec0 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc43706e0d2c4873bfad8f0303068537 : SUCCESS in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74f9b65eabf149d1a8decfde59bdd35f : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ae8a361b034e47b631c32a2ba0b43b : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/91d72533e1914065b05a682464c6040e : SUCCESS in 51m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9240cb0c8fae47bbafc4e192ad43d561 : SUCCESS in 1h 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d377ff7aa8c4f23b8ed3d966a070f5d : SUCCESS in 57m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af3a7dd289442d1b1d71d950f6dd368 : SUCCESS in 18m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/56dde95259f045d5a46e41ca30ce2366 : SUCCESS in 19m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a949d7a262f14e0b8e053b4e4bf12e9d : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e112f348658403c899d8701cddaeaf2 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00f9d9cdf88b421f856b58dff5408311 : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0515d1117b714f97a0cf35d6949670b8 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a1ea04f88694605926653e4f8bd4c66 : FAILURE in 39m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ac967c680f1402ca267f2f0d682a0c2 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5b8323791d34309a52dae379e2b5e42 : SUCCESS in 1h 25m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50882b8f1208401e9797ddcac6a23416 : SUCCESS in 2h 29m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92b483c46d8649cca093aeb2e31f1233 : SUCCESS in 1h 47m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2272e158bfa54d1786cf0bdf7e9598c8 : FAILURE in 8m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/909c256d7000400c9b7bcbba7ec3fe56 : SUCCESS in 30m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d412e59a6e843b8bf94fd8f33b05db9 : SUCCESS in 1h 04m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac33d48f82f04c5b952930652109ec69 : SUCCESS in 1h 37m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a5b5e31eae8479c8c3a4758bc454e19 : SUCCESS in 39m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d294fed48d294813adafb6ffbe4c9de0 : SUCCESS in 1h 20m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ee25dcebcee04891b7f31a89829aafab : FAILURE in 7m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c9286df933645b58da5d842f29dae15 : SUCCESS in 1h 14m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/888ba01eaf0c4266a917b9bde0874490 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":8},{"id":"ec87c75358eed352dd5650d1bcc8ae7e6cdddb5e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-05 12:13:43.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_08_2024_12_07_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"0470bc52aefa449a6c6ab7c9bbb5af3c05803fe0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-05 12:26:33.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/da2ebe794e5248f4ac5441e544bf35e1 : FAILURE in 1h 24m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9c25305d524d4093b6c7b80b58d836af : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":8},{"id":"b38f13dbc6045d56ea3467ac98f98f585771333e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-05 12:41:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/924505/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/14e2b43/ : RETRY_LIMIT in 8m 38s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"ddde6e9961fc67443ccd190189b34ab4bd6098da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-05 12:54:55.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/843/.","accounts_in_message":[],"_revision_number":8},{"id":"712b638249116d626d7c85fcff5174c75f1b0c87","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-05 15:28:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerStore_iSCSI/72caf23/DellEMC_PowerStore_iSCSI/1569 : SUCCESS in 1h 17m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerStore_FC/3ea6d80/DellEMC_PowerStore_FC/1764 : SUCCESS in 1h 16m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerStore_NFS/0c311ed/DellEMC_PowerStore_NFS/1262 : FAILURE in 1h 12m 52s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerStore_NVMeTCP/b45e287/DellEMC_PowerStore_NVMeTCP/1121 : SUCCESS in 2h 02m 17s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerStore_NVMeFC/f3d44b9/DellEMC_PowerStore_NVMeFC/85 : SUCCESS in 1h 19m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"e297724e3fae3a3d622a8cfe4611a7e672745a63","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-05 17:24:01.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_3.6/2e641e6/DellEMC_PowerFlex_3.6/1843 : SUCCESS in 2h 01m 46s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_4.5/6af037d/DellEMC_PowerFlex_4.5/722 : SUCCESS in 1h 48m 36s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/1e6d937/DellEMC_PowerFlex_4.x_NVMeTCP/686 : FAILURE in 12m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"a12be501bf154d8a94f9b820c023ffc023c05395","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-05 17:30:10.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_Unity_FC/e903160/DellEMC_Unity_FC/2357 : SUCCESS in 1h 29m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/8/check/DellEMC_Unity_iSCSI/9d884d0/DellEMC_Unity_iSCSI/2362 : SUCCESS in 1h 57m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"20a4a9e0d920132ff8004ba97af955eaedb52173","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-05 19:15:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/8/20240806000903/NEC-V-FC : SUCCESS in 3h 37m 48s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/8/20240806000903/NEC-V-ISCSI : SUCCESS in 4h 06m 52s","accounts_in_message":[],"_revision_number":8},{"id":"79ba4b98819ef5d6f18cf680255a7bf7e755f46a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-06 06:08:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/8/20240806092309/NEC-ISCSI : SUCCESS in 5h 17m 44s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/924505/8/20240806092309/NEC-FC : SUCCESS in 5h 45m 18s","accounts_in_message":[],"_revision_number":8},{"id":"d358a446f85ab5efbcdc3e4f1ddbe20cb6f8054f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-01 12:50:48.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0f40b61161b3a83a5b729e2cabbad55cba645a0e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-07 04:22:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2647e2047f4c5eddb2bcdf562cb4da96d8e98c3e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-14 10:24:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8294da2b9d8ff030e42bf845af38fb349a5806fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 09:56:08.000000000","message":"Topic hpn removed","accounts_in_message":[],"_revision_number":8},{"id":"c84d1191926f929e4c519b02573e5e636ee80197","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 10:02:13.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* cinder/tests/unit/volume/drivers/hpe/test_nimble.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9799f26654918f6b567129771bcf650ab14e286c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-15 10:02:43.000000000","message":"Patch Set 9:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":9},{"id":"cf99f1b3bffc5f230f12e5b32c560ce6aa1edc33","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 10:11:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ebce9d3d70b18204bdd8426d4d197edc9d159aae","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-15 10:11:31.000000000","message":"Patch Set 10:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":10},{"id":"999378d8f5e7755185e87760b1f776912fa92f94","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 10:22:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5a0e7639282769a566b33fbc826260b0a194fab5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-15 10:22:58.000000000","message":"Patch Set 11:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":11},{"id":"38729158e44cc3853c1a6ac9880f003dcb8685f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 10:24:18.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/hpe/nimble.py\n","accounts_in_message":[],"_revision_number":12},{"id":"b068b1650a57702f66de518b7050d9abc8776bb8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-15 10:24:39.000000000","message":"Patch Set 12:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":12},{"id":"b2ceaa2c16a9b95582476ad2d02e0bed8dedcc27","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-15 10:46:57.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 21m 01s","accounts_in_message":[],"_revision_number":12},{"id":"3add931aeaf63558c0ae479b99b2fb6c802e70db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 10:48:31.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7b3c5d4548410399726664fb93989b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6e0c7acc9694b5582f0050fa5c70bf9 : FAILURE in 11m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/504474f718f847638b1b54ccb2c5d39e : FAILURE in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"05a2747c9050a6fc5c7246a571b8db68102fcf91","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-15 11:35:17.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924505 : FAILURE in 1h 01m 27s","accounts_in_message":[],"_revision_number":12},{"id":"45c42ecba6a1b6a100d5262a03fa317bd1870905","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:04.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/740/","accounts_in_message":[],"_revision_number":9},{"id":"2acbaafd190010203e40cac38b9004b9b528ceef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:06.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"2f24c27e607756c992345e6ccec17891a695f789","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:07.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/741/","accounts_in_message":[],"_revision_number":10},{"id":"3b97b05f85d5cb98f968d9cff0c3030bae09f6c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:10.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"ff853b9934bbd41e3e963b73924c92f0367c5889","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:11.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/742/","accounts_in_message":[],"_revision_number":11},{"id":"578e9a7810cb1ae108c5a3e86cfe437683b727c5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:13.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"a72b159beeb2a2ceabf8660e99b4a76f459db664","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:14.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/743/","accounts_in_message":[],"_revision_number":12},{"id":"b043a1f0f6aa541fa37357e86a8fe553d7a0c432","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 11:51:16.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"ba3c5422f7b926c158d8a7c7593a8fd59e9609ba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-15 12:21:19.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d1c924d4df1a46329f4ed4c33eeaf0ef\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2b3a27609fa4e14a56db867163a4b5f : SUCCESS in 36m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07e130edccf345af967ed13a1b939f91 : FAILURE in 37m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c93e8b6140c4c23b09f264f31d7c0a7 : FAILURE in 1h 51m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c769c6dbdaf4629a0e5309c917bc724 : SUCCESS in 41m 49s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30bbb37850a346a8af5518ede120e645 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7fb73855df9d4633aac1b7b14bd2ee55 : NODE_FAILURE Node request 200-0000010240 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2243893d3d0e4e0cb0b75813a842025e : RETRY_LIMIT in 4m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d218fd4d62e7431898e648d5f99849bd : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"d800a074382bef0a0ec4cec5e5b18fafa440aa56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 12:28:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3abade496afda1b5048ba1583287cb8c95438ea3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 12:28:41.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/744/","accounts_in_message":[],"_revision_number":13},{"id":"513fb8227e5569a84909499b06f15a8ba043256f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-15 12:28:51.000000000","message":"Patch Set 13:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":13},{"id":"5f78019d7ee7cb85aed46cb778f1608cd98e2f69","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 12:29:16.000000000","message":"Topic If8d91a789dd7619404ca493fda3cd230fdecc651 removed","accounts_in_message":[],"_revision_number":13},{"id":"4f8c27d08ef01454aed6be934cac8e74d9877606","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-15 12:35:28.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/13/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"628fab097ffe80375f997acea4cef2064cb70b7a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-15 12:46:03.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"39aeb137c521f8f42587218fa6ce7e82b54bb9e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 13:11:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531d1d1bbc7b429ea808d70e4577f7cc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c3e93e89464414f84d0344a73a0257e : FAILURE in 21m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b57f86a5d5954908bb92c7044f13eda8 : FAILURE in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d6ddf930ddf4000a6b737adb7b67cec286c4493a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-15 13:33:23.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":13},{"id":"4cc2c201b37f8901d49db38602a2843a1d029a84","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-15 14:32:54.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9b7911289e54189a0da5d6dbd3648ea\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/301093b5171a4126883eb7489c2b0755 : SUCCESS in 45m 36s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce09e91254c0412a89a70a1d7c9e53ef : FAILURE in 46m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3656a520b0e648139d1d45129a58ab2a : FAILURE in 2h 03m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2c34b2d20f249829a3eb5ce0a01cf1e : SUCCESS in 34m 10s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03e7377f66a148dabb800ffc2799d325 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/256d785f215b4bc1b0ee3d43757bc5b8 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a6ef22c10e641a7bdeccfa78477f51e : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88720345f8634ed8a3c6bb1a4e958b86 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"e81679c493c8d6c08d22cbef7142ad220bc04ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:00:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15af58beef04179990a0f1a1eb2db83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/040e3b501e0545c58b5c110d21e0bc14 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ae3f0ca9644461ba62bba0326c91aa2 : FAILURE in 11m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/548c6b3765ec45d39694d2c68fb2a6bc : FAILURE in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477d95e5812e4a298d0a1964fdeb2fbb : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/300f0bd23ab844e89fd75c6f4db66364 : SUCCESS in 44m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c6733bcb29b4d82b932ca870c280c23 : SUCCESS in 48m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d45a6cff6462429280d4fc780f364e24 : SUCCESS in 1h 41m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a18ccd50517649ab8ed05e45bde77780 : SUCCESS in 56m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b1d966e813746ca87d8b34613a0d9d7 : SUCCESS in 15m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f8fe457cd51478db6b045ffee21582c : FAILURE in 13m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/81900d112b55475281a729ac43dbec4c : FAILURE in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e3991bf8c7443b1b98f0fa45f7ba76a : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9a4c136a90d468fb871c168d1c95663 : SUCCESS in 5m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a9626d16f49b43d48b9e9dc323cad1dd : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/42d8870feac6410c97eb1a5080f6b678 : FAILURE in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d03901b2e50e40428c870cf7a607b3f5 : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b2de36d3835641dcbfab335ca1e5ae39 : SUCCESS in 1h 46m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/deb32d9336c14b35a7a5eded9c0ec300 : FAILURE in 2h 25m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d87899f0bf1d4f49b51305d7008ec1e5 : SUCCESS in 1h 35m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/29300f646c2e404982efe8e1da2213ab : SUCCESS in 1h 59m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/765434fa8bf743aca881118573a268e3 : SUCCESS in 38m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a534ed738f4432486f2e8b9e1249b63 : SUCCESS in 1h 10m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f34196b91b0d4eb180188ec8ead9b113 : SUCCESS in 1h 31m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a080869212d4c83b2c740d3abd32523 : SUCCESS in 38m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1ac3dabb8d94829b86851ff3c3f2870 : SUCCESS in 1h 17m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/075b1c85db974df39bae7272f871c8bd : SUCCESS in 1h 29m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d8f8c18353d498ea8f129f450075abc : SUCCESS in 1h 53m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d83a7a23f13943318361497c15e01fda : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":13},{"id":"e883a5de2e1fb93616db49b70fad2eec2d648ff8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-15 15:30:35.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924505 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":13},{"id":"8c465dadd3672011d81727b126cf5df93d31869d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-15 15:56:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8a67c3b71004ffba6dd10fde8a3e2e1 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fb0d7171d27243f7ac687b6b02dd77b0 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"92cc019e1bf0886503b85c2414effb2b3f5d9e06","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-15 19:58:58.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924505 : FAILURE in 47m 41s","accounts_in_message":[],"_revision_number":13},{"id":"3a8f17485fea54183d90a8437a78f4dc7e183b99","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-16 03:05:20.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/13/check/DellEMC_PowerFlex_3.6/6d4c239/DellEMC_PowerFlex_3.6/2241 : SUCCESS in 2h 29m 12s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/05/924505/13/check/DellEMC_PowerFlex_4.x/3012b9e/DellEMC_PowerFlex_4.x/1124 : SUCCESS in 2h 26m 35s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/13/check/DellEMC_PowerFlex_4.x_NVMeTCP/53ecd4f/DellEMC_PowerFlex_4.x_NVMeTCP/1049 : FAILURE in 5h 54m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"fd14b80e5edfe359d0c01650d46d90e13a64f5ed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-16 04:50:19.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/13/check/DellEMC_Unity_FC/39a84cd/DellEMC_Unity_FC/2742 : ABORTED in 3h 00m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/13/check/DellEMC_Unity_iSCSI/1b9416c/DellEMC_Unity_iSCSI/2737 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"4c762a5b845e55c915aa7b9befb0b6826ff893fb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-16 19:36:56.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/13/20241016225113/NEC-ISCSI : SUCCESS in 5h 37m 32s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/13/20241016225113/NEC-FC : SUCCESS in 5h 45m 39s","accounts_in_message":[],"_revision_number":13},{"id":"8e5ca588d2b8e7ff107a5f45d1f175f1f927fef3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-18 10:02:18.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"dd7f336eb0c41ec038b30d84481f29e89e903d3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 10:02:29.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/763/","accounts_in_message":[],"_revision_number":14},{"id":"233b92c8983b5077fa31b54e41fc76eabb1cb5f4","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-18 10:02:39.000000000","message":"Patch Set 14:\n\nUnable to freeze job graph: Job fujitsu-eternus-dx-fc does not specify a run playbook","accounts_in_message":[],"_revision_number":14},{"id":"a2e1e87f899e35acc0e17bbf0b43d4f6292bc89c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-18 10:04:55.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d2d512cbc18effc920c08cf8b51fd3ea7fa55c98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-18 10:09:36.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/14/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"25ffd147e4acc8aec4346484b7803efb29899814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 10:32:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb12493c3cb452da2d01c59fa62f33e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7f754c8a0fb4ff482495196d81e2664 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26d9b93d9e344c87897c5ce4927e3a57 : FAILURE in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ba24d3a7d4d3bdb165d22d18faf4c5adcaba50df","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-18 10:37:15.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924505 : FAILURE in 22m 46s","accounts_in_message":[],"_revision_number":14},{"id":"2360f6a31f13d4c833d2239825477ce0abdb7481","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-18 10:37:52.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":14},{"id":"b6fc6d1dede0edfba45885539df62f92d2377ebf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-18 11:14:17.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/715/.","accounts_in_message":[],"_revision_number":14},{"id":"4c35db0635a4012c2136698e865d897cbcb47137","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-18 11:19:44.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924505 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":14},{"id":"ff39c32f290ad1ea4ebc63aa6f195e84e37871df","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-18 11:52:50.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1289/.","accounts_in_message":[],"_revision_number":14},{"id":"6dac6824679a77019e6f8df6b2c0df210fdbc999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 12:46:51.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d107e2b3d0240bda75d5c45cc8eb49c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25575033c504cceba6cd1b10266a488 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be156d6642c94aa99ab4aef0cb0112ad : SUCCESS in 10m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e684b22dbd2d4315aaff56bb3cdb51b2 : FAILURE in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912d43bea69a44bc93797b7a2dc97f1f : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/37c17330fa81433889435c431734106e : SUCCESS in 55m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e3af96b8428400f987888fe628c7b25 : SUCCESS in 54m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab41500d64464de7b66499f70996d6bd : SUCCESS in 1h 31m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e706ab15ce744340975360e96bbd93a6 : SUCCESS in 55m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3da7a171818c443db3b3fef0581da382 : SUCCESS in 16m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ffadfd304f0343089435b7cbb1f16a61 : FAILURE in 12m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fd61c40ec3c94a1a95a7d75edf9a111b : FAILURE in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6603369dd3a4c2095acd5c240718aec : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65766c122a5c4aa7a02369fbced50dd0 : SUCCESS in 5m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/624ba65a3a2b4966a17c63985c821da3 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ef935bf4b3146319bb1840dd2443685 : FAILURE in 30m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a87425ac9c94bd5aa0a43db9b4a145f : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b46556d62214491b3dc778b5dc2b37c : SUCCESS in 1h 29m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9bdbf80961984bcea548aa4c779bf106 : FAILURE in 2h 28m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5eec1528ce2b4a0f95844a6793a8b3b9 : SUCCESS in 1h 38m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2dad9f65f54c4b04aaa9c66777c78108 : SUCCESS in 2h 09m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e8d5cef944e4809b1ca43825abaf67e : SUCCESS in 29m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba801e9fcb5f47deb14fac6dbfcfa1b3 : SUCCESS in 55m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3951c25cc5d34a71a12f0e2f142148cf : SUCCESS in 1h 29m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc8687cb76f245e3aa794d6722fcabae : SUCCESS in 44m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3e6cc7be0fc4ee1b5555c4edbaf8aad : SUCCESS in 1h 30m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/462b4ff3860645368e2f0f727c53b4ee : SUCCESS in 1h 27m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/527bd98079724197a6b2f397eb58626a : SUCCESS in 1h 25m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a93f875365134f33999ec903a7a478e3 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":14},{"id":"cb322d5d6a8a73a47ed3836fff1325bc75166cc8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-18 13:22:12.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62b28ec2d3fd4243a355ec38ea11d69b : SUCCESS in 1h 14m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b13c5b32b3234131b09c0acdbd7ed95a : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":14},{"id":"2fa93a97b4b093184a87697cb181a794600326f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-18 14:41:04.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/030ec68e740d4b309a3eee76a03a99fb\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f9237707e2bd4dd5b09460cfc9869851 : SUCCESS in 43m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac11aaa836d4432196d80cabb3ec0498 : FAILURE in 45m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/643d6d4ca84949c2a2d4b7dd6c972c63 : FAILURE in 1h 29m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6ba011526134be0aba79edcfcdf396a : SUCCESS in 45m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a846c49957104c5290ce7b9a1bc2fca1 : NODE_FAILURE Node request 200-0000010542 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d946138edb6048e5abf0b1c420ecc52b : NODE_FAILURE Node request 200-0000010582 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c3e483546aa7468889c3b6004f3dfad1 : NODE_FAILURE Node request 200-0000010696 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5fa45f4c27e1460581479b84ffdce813 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Var... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"c6da6c8ea233fd113ef60e35feaf3c0d28a7f59f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-18 18:31:15.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_3.6/fa640a0/DellEMC_PowerFlex_3.6/2255 : FAILURE in 2h 33m 19s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_4.x/70e90dc/DellEMC_PowerFlex_4.x/1139 : SUCCESS in 1h 52m 04s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/4ada0b9/DellEMC_PowerFlex_4.x_NVMeTCP/1060 : FAILURE in 5h 52m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"9b0621ace90880d17099a74a77485c6d639f0b1d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-18 19:51:27.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_Unity_FC/71af69c/DellEMC_Unity_FC/2756 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_Unity_iSCSI/d08f632/DellEMC_Unity_iSCSI/2747 : ABORTED in 6h 00m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"cc9d0879bdd3abc9b64876cf106c4da7114c31e8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-19 05:50:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/14/20241019105655/NEC-V-FC : SUCCESS in 3h 37m 33s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/14/20241019105655/NEC-V-ISCSI : SUCCESS in 3h 53m 26s","accounts_in_message":[],"_revision_number":14},{"id":"582e064a49eb52e75bbf424b69b9d6e400a523a2","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-19 06:18:45.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c6a3c99ab8f79a5f11448c375fafaffdec355569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 06:51:51.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df96f777c7fa4867bb1139c6045fbcf7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c534004f4ff4bd9beb5f129454e6ec3 : FAILURE in 29m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2d9c12323fe402ba7119b91fc9bc387 : FAILURE in 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e53472e90b57cf8a9d4666c02793812c5fbe3eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 08:45:28.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9982d40edcf45bbbf2eb1eb7d918b35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538aba176c1c41a097767318a9fcf3b2 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49fdc71023eb4028b0038f4818ae6d2c : FAILURE in 12m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad5a2157323e4f7391eaa87e9b05290f : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bea9c9560a1434cbc6c05d69c5d89ef : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/67e0cfa287d44a8aaa8939c36f62811b : SUCCESS in 1h 01m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf7ad8d1a0544b39a8ea2a8ad9c1acbc : SUCCESS in 47m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b18ee5b638844ef9905a69d4a21eb17a : SUCCESS in 1h 49m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc575f707d8c42219df5287dbaff8186 : SUCCESS in 1h 05m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78db9f95cf574c17a8bede784a489aff : SUCCESS in 19m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/864347008b1f43f097e8ac073e500a53 : FAILURE in 14m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7163130a45d400cb65f6c1202b54092 : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a3016a2e939475eacc9c290205af83e : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13a76b99119044cab5b5dc02905e8d89 : SUCCESS in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e3ce82147576401ea149378a0beada8d : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/213e72ddfd24453faea0436df2fe4660 : FAILURE in 33m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81d014d1563c44b89d76933505232713 : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aedc475c1bd54c308e240252e112c570 : SUCCESS in 1h 07m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c77943227ee54da2bbc04d0fbd35c2ce : SUCCESS in 2h 09m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90fe7d1f1ae143ebbda2e79d2c40d8b8 : SUCCESS in 1h 27m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/511b5f4866ee40ea9343cce197d99cc8 : SUCCESS in 1h 56m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/22641ded44334dadb2471b78e789a303 : SUCCESS in 23m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91749525397f44e49219599e6531502c : SUCCESS in 1h 13m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/982775a6c7e24e2bbda94bbcae27b3ce : SUCCESS in 1h 13m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bafb740c7e9d41f3b46f8af161767241 : SUCCESS in 31m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a59bc9d880a4bfda0a1c50ced389174 : SUCCESS in 58m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/079be18e8bdc442a8ec1c3f61dcf500b : SUCCESS in 1h 51m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae208ae4a0b8409484527cd2c385f061 : SUCCESS in 1h 50m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a37f7758cb364d5f86ec285996935057 : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":14},{"id":"cc7a129231dddf1a15f6e0e68fbe4643cda806d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-19 10:15:52.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/66499c65e0f74117a2c6f2ddcc1627c9 : FAILURE in 1h 20m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f4752c4f9e245adbc353d6124b05e04 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":14},{"id":"103df41014ca8aff242732b77bd7b561c6da6ea6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-19 14:59:48.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/14/20241019155519/NEC-FC : SUCCESS in 5h 40m 48s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/924505/14/20241019155519/NEC-ISCSI : SUCCESS in 8h 04m 25s","accounts_in_message":[],"_revision_number":14},{"id":"2bfb29d633bb720b40ab7680de8ea53461672620","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-20 02:53:03.000000000","message":"Patch Set 14:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":14},{"id":"7b8279b2d031e6db40ba3279a35a2c81ad095221","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-21 04:20:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e075e9d0969f376f20f88a31d46d07ad594ab774","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-21 04:22:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"16dbd022c6db74f41cdb07ad620044e3c813b3cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 04:43:20.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3288fcddd20940f78c5325779d8dc3c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e79b7ec327e748959f41ee01385f9125 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/039e9b69fd4a4736ab72e7324774d1ff : SUCCESS in 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e90e901a9a62ae57b3422530bf837ff99883dc75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 06:45:27.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd9602fa133046018cf27b1713eab46c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a526476042344104bb86d525c304195e : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10554284028c431297f593522da1fc14 : FAILURE in 8m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be8c5d1576d8467eaaf733e50d202504 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b492b380d884d28ab8c2ed25d961f56 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a1d2db7aa52b41daaf93f8ccf2db7274 : SUCCESS in 1h 01m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c257fa31eb843e3bdf6a7ab2b56d558 : SUCCESS in 44m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e063034e6e2c47738056eb53bbe266ff : SUCCESS in 1h 23m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b42df551bd44c58a703dff1212f551c : SUCCESS in 44m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d2865680564cd1832b9037a83d0340 : SUCCESS in 17m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d3820ffb6004992b6e4d7d31a4fbf21 : SUCCESS in 14m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2cbc2bb4e0fb4d53927eb8a391938b56 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f30bdb5e6ef64ecfa212a06cc4141b39 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b61e16a6785743babef5800496e96377 : SUCCESS in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/42e3a2b755c642c9a56e0cad653be1d1 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca7570e832e142c79b1efa8745ebf054 : FAILURE in 18m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77c2374ee4cf4e55a26f50b370e0af6c : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/12333b900dc544ffa2f5f999c34a272a : SUCCESS in 1h 40m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d8369f727dd4f02bebd7e203e8af54c : FAILURE in 2h 07m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6e1c6936c60042fe91b0eff2584d44bf : SUCCESS in 1h 28m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/119754ae1e094575a6090e40427fc45d : SUCCESS in 2h 17m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1c92660297c64eb5b6d99216030fc6e5 : SUCCESS in 24m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ecafe6939d442bb9eb42ad289da384c : SUCCESS in 45m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5dcfabb3ac4e4aaeb948011bd22c6930 : SUCCESS in 1h 30m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ad18245ba19143f5aaed9cde48590f70 : SUCCESS in 42m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f74116267524ced93bd5eb8a8f3925b : SUCCESS in 1h 03m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5169972c4e0743f2a9ab87f04b4d74d1 : SUCCESS in 1h 47m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b894264c6f6481aadbaccfb02af6e4c : SUCCESS in 1h 08m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/278cca98740c42d2b7f083716f22951b : SUCCESS in 50m 47s","accounts_in_message":[],"_revision_number":14},{"id":"01f846552be50f8871462633eb18617fed2549f4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-21 06:51:29.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6471d802e365a288f91e03b293211032345794ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-21 06:59:54.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6a20ed1f714a47b098c60f1c3412554e : FAILURE in 1h 09m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e79caff0e364415a89502e24d43a6b4f : FAILURE in 1h 17m 08s","accounts_in_message":[],"_revision_number":14},{"id":"a45398f2c9c4e72ba8182ea7fcd00c9ece6c2e02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 07:12:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8082e6489db04c29aec2a0d34cdbc353\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dbcf4eb3b974e58854f8a21d556e7af : FAILURE in 19m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8d93223a0b746f79f153678b9118dd7 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4acd97ea27ce95538b2d97ed35e958a5ee754a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 08:57:09.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34d5966b2f5b4f7dbff893a5d9740203\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb87986132fc4097ab6d2a69b774ed6a : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5ac7cf274244569a1cd4402a1842faa : SUCCESS in 9m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5941552cb097449493d59b90f5d3b5b2 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf55f528a421437db2dcc9fc52c85225 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/275a8864937d461b926342b6e1b344af : SUCCESS in 41m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dde6f44cb7f4bad8757c14f464698a2 : SUCCESS in 42m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fe246b192634653b0687c91f464cfe1 : SUCCESS in 1h 31m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f030765587704494bf3801e65a5a574e : SUCCESS in 1h 06m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a7249a04d734b31b0ad9a0c9b6e56aa : SUCCESS in 18m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0ab84f2576324dae864cb0822048be6c : SUCCESS in 13m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a54ef4795eb644ba8507d9a9354b30e8 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b9d547deda44f6d844868e8d0977b4e : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e56a7e9354c4e039f5aab0c31bfcd81 : SUCCESS in 5m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e8a3ecbb4ec64ec9b16ca21de1eb8bd7 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c44412cf11e48b28e00415fa6b56c09 : FAILURE in 32m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba9a93efe41e40a9b45c069c9e73c7b3 : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4677b5ff012c44d4974672e3341131a6 : SUCCESS in 1h 44m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8184b7c903140059e237eab44acb63a : FAILURE in 1h 32m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3df326d33dcc44c581df1b808b3cf51e : FAILURE in 2h 04m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/819b143b4af848c3bbc64f9d235ce45a : SUCCESS in 1h 27m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f16f905858341b9b787e89a1b591b01 : SUCCESS in 31m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/209223f839ba4012b1be0487788d7fb3 : SUCCESS in 50m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5604a40e53374c65b3312cb81c4f10a1 : SUCCESS in 1h 43m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a89979338864815bb57efdb6f07053f : SUCCESS in 42m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e3c17f2280c4f05a26d9ab0128c82f0 : SUCCESS in 1h 01m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fc35992cef24eb88cc9c88f05602118 : SUCCESS in 1h 53m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a08103fa4dd24ad3b747686652f7ec91 : SUCCESS in 1h 12m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5755175b931f4cd69e72abcd610782b2 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":14},{"id":"add571bfdbe7aa85123cd2f3308836293163e3ef","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-21 10:33:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f9c66743c626a3f75792fe71bf1ac6ee206fb6c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 10:52:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7042931732754c7e9c55032adf356a0f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a141245f76c14e9a996534579f13974e : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cd82fb62f264830a1a96e2e2dd52ab8 : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fffc7197cc9c33dee28a955c62e4aa6892930007","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-21 12:07:33.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4259f566d2d94caa8ad1d5de5f889da8 : SUCCESS in 1h 14m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2756812e5a3448bc8af976132f6d3928 : FAILURE in 1h 32m 05s","accounts_in_message":[],"_revision_number":14},{"id":"00995b7eb99e8a76de3d0f05d0c7c688c76e63cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 13:05:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226198f6d7f84a88aa72b00f269976db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d15edd72b14c5c8521af71db715397 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d3254e3ca04cc4bbb8b8ef762fd54f : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c2cbabad17f4b168171606bf34bb5da : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867b2cfdb8994caf9a0f7601a880f563 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b1ac324c8cc644b4bbaeec1a4557eac6 : SUCCESS in 52m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf49e14194f3430e9e9324a10c4bc289 : SUCCESS in 1h 03m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4ebcb822e664131b9b8841402836032 : SUCCESS in 1h 33m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08c65e6f496c4abea8c82f88acacee6e : SUCCESS in 37m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50fef2235264b5295e470afa298006d : SUCCESS in 19m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3b2239cf01842e1b1dbcb71a0f79993 : FAILURE in 12m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a65f1b32e4c948bd9dfc7bb7acbe5628 : SUCCESS in 3m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ddd6c59d2e7449e5bcb35fddca283ffd : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/959b2f4eaf6d4f288385787747f962ba : SUCCESS in 5m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16633749bdb049a9a54239c3185f8026 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33f37cb4a0f6471c96b5a9548c54bf65 : FAILURE in 18m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd3232598a4a45aaaad53c656c1303c2 : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ee009b60d5a462baf6ffd825822f809 : SUCCESS in 1h 49m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c6ead650f833483ca5edfa211933156a : SUCCESS in 2h 01m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af155991e27e4d898ce1c9e0129f6648 : SUCCESS in 2h 00m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d94d57e07b1f403baaa236dbff44ace8 : SUCCESS in 1h 48m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6740bde902e44773a795e4237e607cab : SUCCESS in 37m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/168012ecf15d409da70614892a1528f4 : SUCCESS in 1h 21m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ea4a9d2efac492cbe8506ee7519bc0b : SUCCESS in 1h 44m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/28a0edf3c4884658a91e41217c15a892 : SUCCESS in 33m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a49d31c351d6456686572d9e537e72b8 : SUCCESS in 1h 30m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b1e64ad2091b44fdb326ed16f13a473e : POST_FAILURE in 2h 25m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d70e8f5b26f44ba8985a2cb2853c4fec : SUCCESS in 1h 34m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53cba93b742f433396354b3c4dc6b49a : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":14},{"id":"2fc7067ed383a212dbb707f68c49ab225568e6cb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-21 13:22:32.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/eeb5a940590443f0a1663aaa82aff37c : RETRY_LIMIT in 5m 22s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"ea64d1b46b8dd3d6755170704776af2c26b2db33","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-22 17:00:27.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_3.6/839a372/DellEMC_PowerFlex_3.6/2272 : FAILURE in 2h 38m 23s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_4.x/bac04b8/DellEMC_PowerFlex_4.x/1155 : FAILURE in 2h 12m 49s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_PowerFlex_4.x_NVMeTCP/12eee71/DellEMC_PowerFlex_4.x_NVMeTCP/1076 : FAILURE in 5h 54m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"fbd8b1441e896e2e08eb7709c1822e9aeaec3e8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-22 19:53:20.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_Unity_FC/080efc0/DellEMC_Unity_FC/2772 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/14/check/DellEMC_Unity_iSCSI/f9069b7/DellEMC_Unity_iSCSI/2763 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"7b4ffb6d531c166788b28b60b325cd0cd0382ac0","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-23 01:31:38.000000000","message":"Patch Set 14:\n\nBuild failed.","accounts_in_message":[],"_revision_number":14},{"id":"068072a6712c4d68716a5f2c923aa3636838315a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-23 01:49:48.000000000","message":"Patch Set 14:\n\nBuild failed.","accounts_in_message":[],"_revision_number":14},{"id":"aee2a6addb18355ec1cea16ceb517f963629f627","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-23 04:37:02.000000000","message":"Patch Set 14:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/924505-14/HPMSA-iSCSI/ : FAILURE in 3:27:16\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/924505-14/HPMSA-FC/ : FAILURE in 2:38:29","accounts_in_message":[],"_revision_number":14},{"id":"3aaaf563d706db77b2ca0d0247a21548bf96fb76","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-14 14:37:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"591b4d485468c34f6b5a5f7e05947645de83f4c0","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-11-14 18:24:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"759b720547fbd57a33988b303f997bafec432c74","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-14 19:00:08.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/14/fc : SUCCESS in 1h 06m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/14/iscsi : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":14},{"id":"d6a41d3dafde7598c8bdf52af891828de8935327","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-11-14 22:54:08.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b4d6a089e6203e268aaf2c956126c4651c76a15b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 04:23:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"345938bcc3cfaa7e74e61c89cc0359e14721439f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 06:59:52.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"0ca652657fcbe311a847fd0bb1f5c450b3ce935c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-15 07:00:01.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/929/","accounts_in_message":[],"_revision_number":15},{"id":"6cc7bee94d05a8a57152354eace2e0f8beea3489","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-15 07:00:57.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/924505/15/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"77f2f45ade815e1b2caf163e23d021f1f19e366d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 07:03:01.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"fa0ecc249e14db824bcdf388be21f15865f2d692","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 07:17:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c73bb8ecacc439981f98bafb5c4c7f7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/731ec904a7224b739a17b5e079fb20fb : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79935e4f83d246c18312476b33cf7839 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b72c807f1d3c46c6b44c720b2d16a3ba3a2698f7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-15 07:55:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/868/.","accounts_in_message":[],"_revision_number":15},{"id":"6885f70288a40fbd24056fb717bbdc3aa6b80fd0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-15 08:23:36.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/96ca2e7ac583424c86313f67e5d2f1a4 : FAILURE in 1h 17m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5e3d06df5dd14f3180eeebaeaeee31ed : SUCCESS in 1h 21m 39s","accounts_in_message":[],"_revision_number":15},{"id":"33ac5f1901cce6cf74fb63aee262e6c05ce5249e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-15 08:24:03.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5c2a0625ae904d14bfcea35f4cba44a9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3ac5476785564cec95c962e78159731f : SUCCESS in 35m 33s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fee8f98b050d4dc19343035708e7d40a : SUCCESS in 35m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e4ed17f10b0e4972b70bacf17c84bfcf : FAILURE in 1h 23m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46ffc8d40cb94fe8a387a567156f69e4 : FAILURE in 38m 20s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25424b80e7f7419b95a7e5ca23c4ffaa : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef6995dc7dd949a7a29ad6f72cd2eac4 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad625d300930468f836fd7f273477d9f : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c80a099fb54d4c72a6134140a1a14450 : RETRY_LIMIT in 30s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-d... (truncated)","accounts_in_message":[],"_revision_number":15},{"id":"de14756e8b322a765d7715be30f4db9b2b7106b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 09:09:15.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23de0eed7c804a8394d338a947f3fade\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e562e23e0714eb0a5464f14146cba51 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90484c6c1df74b9884d7ae591e4f079c : SUCCESS in 9m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8068aa31d1254be78ce764db9de174ec : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155d26a06d064494a878a272c2ccdcef : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/391fccd409df43508043b0fe4cb24972 : SUCCESS in 39m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1b70b115bbd44a3828ac3fdb54078ff : SUCCESS in 36m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cbc0e619d42641738584bd23f7c52dc2 : SUCCESS in 59m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b254ee14615644dc9b36167acc85113d : SUCCESS in 45m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcfbd159e25c43978eaba7443ca51d6d : SUCCESS in 13m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a62013cc5d884dbea66ed7349106a760 : SUCCESS in 14m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2c46fc2f9a024498b6df174ac24d3ffd : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a733859610aa472b8384f143600c018b : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a302114daa054afaaf35bff3521723ef : SUCCESS in 5m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f482f1575ed478a9c1a42e2be6bebdd : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8c501501a234ff88679f7569d0a489f : FAILURE in 17m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26df0c21409d49c5bfc323dd1c169140 : SUCCESS in 7m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2adbf5dd37084ea390d129fc059941b1 : SUCCESS in 1h 48m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/66c1e2a799e24990bf6ffcefe2200313 : FAILURE in 2h 03m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/215442cd26c7482a93d29f14b34f81ea : SUCCESS in 1h 44m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1d9be2e8a68b49c8b732fb253db86a7c : SUCCESS in 1h 28m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6421ba6a6ce74593b0ce69f8677ae3c3 : SUCCESS in 31m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9295fcca1d81407784a388c533e8f5b7 : SUCCESS in 1h 15m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0144888c8aa740e1a2635d35b6984791 : SUCCESS in 1h 38m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f5c5e33725704ab9a0b48dbda91e4c4b : SUCCESS in 30m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/00d0b2f549024b78b00f2d19e7e1b542 : SUCCESS in 1h 16m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e8155601b1874254bcd49393812dc25d : POST_FAILURE in 1h 55m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b8b611535a340908493053f5f6d8b58 : SUCCESS in 1h 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba9fd13d7c0a4dfe9bda4ebe6fd8105e : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":15},{"id":"a3ab00ad5f4abfa2208de88963e87b375a8892a5","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 09:18:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4d4e79a9d21f8bad092030bc36c3ecf268f1e727","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-15 09:21:16.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/17f8e6026e3d4f2b867ed31f563c137c : RETRY_LIMIT in 6m 05s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/0a96c089b66748e2904bd2c82d6ba6cc : RETRY_LIMIT in 10m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"94dad30de36af22e13fd52fd9d422c64f9ede4ce","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 09:45:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ee71b15f8ce88d54919dccdd5d14a191083a1c5d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-15 10:00:08.000000000","message":"Patch Set 15:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/15/fc : SUCCESS in 0h 39m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/05/924505/15/iscsi : SUCCESS in 0h 53m 45s","accounts_in_message":[],"_revision_number":15},{"id":"c3dc8ea66ce554efbcea95e4da9a3fa2ef9e4f99","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-15 10:48:46.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924505 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":15},{"id":"cb3b33887d2bb7cd86fb4327af261b93d51cf658","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 11:45:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bd15d983f7ffaed01d4ac29b2a24ad1b9306af49","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-11-15 12:50:55.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ec506fbdcacc7730fdef79553696f8f7d68bae65","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-15 13:12:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cde22cb4c5df572c724bdb9727984d248ea0ad5d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-11-15 16:21:27.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"241f3be241dadf123ac8a88c266eb736bc7d0771","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:22:35.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"136acb460ff23211b09c4050e2b639da6175d862","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-15 17:12:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1434/.","accounts_in_message":[],"_revision_number":15},{"id":"4db805387cf8f526bab03185dffe009a49860f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:44:08.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b3e17826bc4a1ea988ef1befa72112\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fe36dd296c4db0af6583bd68475ff2 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/481a759ae80045e6bfff1b68d100e109 : SUCCESS in 8m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13d24339780746268fb58d7f36d25319 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12fe73fc128b4199bcf5edb89b48857c : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8c8c331288984772b39e95b3b3dd208e : SUCCESS in 1h 02m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5653fc99e50d49768d112e64bce4c9a0 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6318250dee554445a3bebeddd5f0e365 : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af27d0e6722347d98db62bdd885341ab : SUCCESS in 51m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b64145076042cca44877ca3f4231d8 : SUCCESS in 19m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5acded024d2f4489bc79e5d1647cba79 : SUCCESS in 1h 21m 00s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c33e0ef410df43e6a531e827fc94e575 : SUCCESS in 1h 26m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6e7ca3dbd78400abbed11135bdb1c81 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":15},{"id":"cffa4f27cbf552d0f45e0bde321a7972c4db138d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:44:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"452479f1e1325a822b16d8b03c2efb007799c272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:45:47.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb14081df25e46a1a2b7d68876c7d2fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299c49dee9944f1c889e1b135a526393 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a5575f2ff8441c5a6e53cca0afa1ab3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":15},{"id":"718bcafe34859c98c2703c26f030c51278834bca","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-15 19:20:09.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/924505-15/Seagate-iSCSI/ : SUCCESS in 2:27:58\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/924505-15/Seagate-FC/ : SUCCESS in 2:19:29","accounts_in_message":[],"_revision_number":15},{"id":"28583b7b166283a67e440dcc6ba904e60ac9a936","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-16 10:14:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"001dd53a817f0420c1b2ca8fa7cb2d5a95e3c66b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-16 11:17:46.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/05/924505/15/check/DellEMC_PowerFlex_3.6/f8121cb/DellEMC_PowerFlex_3.6/2435 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/05/924505/15/check/DellEMC_PowerFlex_4.x/b500688/DellEMC_PowerFlex_4.x/1283 : FAILURE in 2h 04m 01s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/05/924505/15/check/DellEMC_PowerFlex_4.x_NVMeTCP/e6f905b/DellEMC_PowerFlex_4.x_NVMeTCP/1176 : FAILURE in 4h 43m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"8c4d51f0c77ce38775aa431e54f3675755028c8e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-16 19:00:32.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/924505/15/check/DellEMC_Unity_FC/c126cd4/DellEMC_Unity_FC/2887 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/924505/15/check/DellEMC_Unity_iSCSI/cadb6f2/DellEMC_Unity_iSCSI/2854 : ABORTED in 6h 00m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"cf76f8e4f48d198cf5733c93ba720a68a94cddeb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 17:06:30.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_17_00_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"05f6b708a30eb6cbeabdf0996128e141386b0ebd","revisions":{"654b3e94f8879ffbd24c6d9930b1478322739878":{"kind":"REWORK","_number":1,"created":"2024-07-19 12:19:01.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/1"}}},"commit":{"parents":[{"commit":"93e5997bdd9b4dc17890b35f9dcb4bc11b2eb38d","subject":"[HP Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93e5997bdd9b4dc17890b35f9dcb4bc11b2eb38d"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:18:12.000000000","tz":180},"subject":"[HP Numble] Report thin provisioning correctly","message":"[HP Numble] Report thin provisioning correctly\n\nReport backend capability according configured value, not\nhardcoded flag.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/654b3e94f8879ffbd24c6d9930b1478322739878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/654b3e94f8879ffbd24c6d9930b1478322739878"}]},"branch":"refs/heads/master"},"c279e0d19da8a007803842b06ce7f8d343875872":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-19 12:20:07.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/2"}}},"commit":{"parents":[{"commit":"93e5997bdd9b4dc17890b35f9dcb4bc11b2eb38d","subject":"[HP Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93e5997bdd9b4dc17890b35f9dcb4bc11b2eb38d"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:19:19.000000000","tz":180},"subject":"[HP Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HP Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HP Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c279e0d19da8a007803842b06ce7f8d343875872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c279e0d19da8a007803842b06ce7f8d343875872"}]},"branch":"refs/heads/master"},"81a110d88cfa8fb2ea19f9026d0130e99d00085c":{"kind":"REWORK","_number":3,"created":"2024-07-19 13:28:58.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/3"}}},"commit":{"parents":[{"commit":"7d12f503d4450da04314052eb1efff7c96c796e8","subject":"[HP Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d12f503d4450da04314052eb1efff7c96c796e8"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 13:28:41.000000000","tz":180},"subject":"[HP Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HP Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HP Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81a110d88cfa8fb2ea19f9026d0130e99d00085c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81a110d88cfa8fb2ea19f9026d0130e99d00085c"}]},"branch":"refs/heads/master"},"0b70449850665aff6eba0f4140e306a4e162c32e":{"kind":"NO_CHANGE","_number":4,"created":"2024-07-26 07:46:47.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/4"}}},"commit":{"parents":[{"commit":"5c10d3b54c640ebddcfe99c668543c1511c82598","subject":"[HPE Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c10d3b54c640ebddcfe99c668543c1511c82598"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-26 07:46:47.000000000","tz":0},"subject":"[HP Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HP Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HP Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b70449850665aff6eba0f4140e306a4e162c32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b70449850665aff6eba0f4140e306a4e162c32e"}]},"branch":"refs/heads/master","description":"Rebase"},"ae2b8476e629d516ccf4553471a184c98e1a9cb4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-07-26 07:46:54.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/5"}}},"commit":{"parents":[{"commit":"5c10d3b54c640ebddcfe99c668543c1511c82598","subject":"[HPE Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c10d3b54c640ebddcfe99c668543c1511c82598"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-26 07:46:54.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HP Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae2b8476e629d516ccf4553471a184c98e1a9cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae2b8476e629d516ccf4553471a184c98e1a9cb4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"70191ffebcf415e95e80e8b11459db12b45bed10":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-07-26 07:47:04.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/6"}}},"commit":{"parents":[{"commit":"5c10d3b54c640ebddcfe99c668543c1511c82598","subject":"[HPE Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c10d3b54c640ebddcfe99c668543c1511c82598"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-26 07:47:04.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70191ffebcf415e95e80e8b11459db12b45bed10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70191ffebcf415e95e80e8b11459db12b45bed10"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bd08ad375dcc1f76e97095e72e2bac1691407504":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-07-26 07:47:51.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/7"}}},"commit":{"parents":[{"commit":"1a008030759ac726f17742df933c597ece420ecf","subject":"[HPE Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a008030759ac726f17742df933c597ece420ecf"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-26 07:47:51.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd08ad375dcc1f76e97095e72e2bac1691407504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd08ad375dcc1f76e97095e72e2bac1691407504"}]},"branch":"refs/heads/master","description":"Rebase"},"bc8559b7e57ef56a3925ce802c28cf0f391b1194":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-08-05 09:25:39.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/05/924505/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/8"}}},"commit":{"parents":[{"commit":"c7c93aad23755d4815b472d7818850e8ab96ef2a","subject":"[HPE Nimble] Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7c93aad23755d4815b472d7818850e8ab96ef2a"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-08-05 09:25:39.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc8559b7e57ef56a3925ce802c28cf0f391b1194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc8559b7e57ef56a3925ce802c28cf0f391b1194"}]},"branch":"refs/heads/master","description":"Rebase"},"ee0372780a173e995788d1ae223275522b05bea1":{"kind":"REWORK","_number":9,"created":"2024-10-15 10:02:13.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/9"}}},"commit":{"parents":[{"commit":"afbbf0f1fecee5a84be6552eb111feee0ce216e6","subject":"HPE Nimble: Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afbbf0f1fecee5a84be6552eb111feee0ce216e6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-15 10:02:13.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee0372780a173e995788d1ae223275522b05bea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee0372780a173e995788d1ae223275522b05bea1"}]},"branch":"refs/heads/master","description":"Rebase"},"d007d4045032f405e23f48400ccc55dbbaf79456":{"kind":"REWORK","_number":10,"created":"2024-10-15 10:11:03.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/10"}}},"commit":{"parents":[{"commit":"afbbf0f1fecee5a84be6552eb111feee0ce216e6","subject":"HPE Nimble: Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afbbf0f1fecee5a84be6552eb111feee0ce216e6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-15 10:27:26.000000000","tz":0},"subject":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities","message":"[HPE Nimble] Report max_oversubscription_rati via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d007d4045032f405e23f48400ccc55dbbaf79456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d007d4045032f405e23f48400ccc55dbbaf79456"}]},"branch":"refs/heads/master"},"fb444edfba7af2c251aafd6e84980d70a52e3f2c":{"kind":"REWORK","_number":11,"created":"2024-10-15 10:22:31.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/11"}}},"commit":{"parents":[{"commit":"afbbf0f1fecee5a84be6552eb111feee0ce216e6","subject":"HPE Nimble: Report thin provisioning correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afbbf0f1fecee5a84be6552eb111feee0ce216e6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-15 10:38:02.000000000","tz":0},"subject":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities","message":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb444edfba7af2c251aafd6e84980d70a52e3f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb444edfba7af2c251aafd6e84980d70a52e3f2c"}]},"branch":"refs/heads/master"},"2238b558bae205361058d7e53afeb1881c7bb7ad":{"kind":"REWORK","_number":12,"created":"2024-10-15 10:24:18.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/12"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-15 10:24:18.000000000","tz":0},"subject":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities","message":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2238b558bae205361058d7e53afeb1881c7bb7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2238b558bae205361058d7e53afeb1881c7bb7ad"}]},"branch":"refs/heads/master","description":"Rebase"},"8cb450556894aa6588a3a4b0696805a94c81225a":{"kind":"REWORK","_number":13,"created":"2024-10-15 12:28:31.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/13"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-15 12:45:32.000000000","tz":0},"subject":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities","message":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb450556894aa6588a3a4b0696805a94c81225a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb450556894aa6588a3a4b0696805a94c81225a"}]},"branch":"refs/heads/master"},"56bf60812fbcc43efbb8298f23b7fe351ce1edbb":{"kind":"REWORK","_number":14,"created":"2024-10-18 10:02:18.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/14"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-10-18 10:18:54.000000000","tz":0},"subject":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities","message":"HPE Nimble: Report max_oversubscription_ratio via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56bf60812fbcc43efbb8298f23b7fe351ce1edbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56bf60812fbcc43efbb8298f23b7fe351ce1edbb"}]},"branch":"refs/heads/master"},"05f6b708a30eb6cbeabdf0996128e141386b0ebd":{"kind":"REWORK","_number":15,"created":"2024-11-15 06:59:52.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/05/924505/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/924505/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/924505/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/924505/15"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-07-19 12:17:46.000000000","tz":180},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2024-11-15 07:13:26.000000000","tz":0},"subject":"Nimble: Report max_oversubscription_ratio via backend capabilities","message":"Nimble: Report max_oversubscription_ratio via backend capabilities\n\nAllow to use different oversubscription ratios per HPE Nimble\nbackend.\n\nChange-Id: If8d91a789dd7619404ca493fda3cd230fdecc651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05f6b708a30eb6cbeabdf0996128e141386b0ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05f6b708a30eb6cbeabdf0996128e141386b0ebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
