)]}'
{"id":"openstack%2Fcinder~924631","triplet_id":"openstack%2Fcinder~master~Id3990df4fbe5914cc3b5e10ccf90aa9e5c1210a5","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-22 18:19:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3990df4fbe5914cc3b5e10ccf90aa9e5c1210a5","subject":"Pass with mypy 1.11.0","status":"MERGED","created":"2024-07-22 13:55:30.000000000","updated":"2024-07-22 18:57:22.000000000","submitted":"2024-07-22 18:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924631","meta_rev_id":"01cacf157ace02fcfb707afe0243ef0e5d28e611","_number":924631,"virtual_id_number":924631,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 18:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-22 14:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-07-22 17:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-07-22 14:04:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 13:55:41.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-22 14:04:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-22 14:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 15:04:00.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-22 15:15:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-22 15:22:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 15:41:08.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-22 16:02:04.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 16:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 16:10:35.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-22 16:14:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-22 16:29:40.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 16:57:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-22 17:05:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-07-22 17:23:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 17:59:15.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-22 18:57:22.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"}],"messages":[{"id":"b2707846c1e793d3eac09aa803f2c17531e7a675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-22 13:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf6a0696cba6858ee478712cbcf09759f399e399","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 13:55:41.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/110/","accounts_in_message":[],"_revision_number":1},{"id":"0a3d9665bdb42d9dc5b7ab5e488adc8d5cd1a09c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 13:55:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/110/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"19de2eee35bf7de179a38afd267e6084b1b27cc3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-22 14:04:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3668328c2e462c293fcf35ea60b24f49a276290","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44fb67de37af4689b48dd0f5ed2d3061\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ce9c714a19342b2a7a0549f87a7f4d3 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09bb1eab1bb64a5287c9b14a66a32206 : SUCCESS in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"340e5c5502d180b72ddd284a13b84a9873637dbb","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-22 15:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924631 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ed67e1a79d02e0559813d2e8f6de957d85da32eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-22 15:15:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/772/.","accounts_in_message":[],"_revision_number":1},{"id":"dc29b5dc54b24ee7fe8828979b524a672774103e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-22 15:22:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924631 : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e5677bcbc16f411e847ebeb64b51adf623a57d39","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-22 15:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924631 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c5e52923c3003c9e43b45d1a773449e68e2cbcbe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-22 16:02:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/90a5c7bb75154c1cb77e90afbc56eea8 : SUCCESS in 1h 15m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1c608b6698994ea2b649b5e7ac2aab03 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cac5c3127b83f0866de389bad5a0a00faab3da7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:04:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a39ae89a62e439bb0c3c709a1afd0cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec68e52d3c854277b9225693a6cf3fae : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd6b2d850eef4952a0ebf59474668cc7 : SUCCESS in 9m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25526f5adef349be95169163bc17300f : SUCCESS in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d705ab4e03e5460ab2d67ac36831d4b8 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d335ccc05f4ae4a85ae8ee91710b53 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/769103f2512347b29dd4d4d45a431504 : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b315894bbab94952914b991d905ac045 : SUCCESS in 1h 49m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/396bef431eb9450eb3ff4295ab84e5d0 : SUCCESS in 50m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/01b2f08ac5294a7d8690f7edf016aa27 : SUCCESS in 17m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b8d4dd23424496da022014db0eed5ef : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/598510f672864c0bb81d79986cb9b5f1 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7c3899e7f43471b829d19cd2990b9ba : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/345c150471514d5aa9b13a30fc94e4b2 : SUCCESS in 7m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/309de9d170cc4a48ac81a1490a3dfc9d : FAILURE in 28m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68218ef00a2e4d0092dca6fac1d8e8ed : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/023838a40ac34aa983f4c2e960f09ce6 : SUCCESS in 1h 32m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6748aab035f84f6897eab183be0c8d46 : SUCCESS in 2h 08m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1dc8fa61443b4fa393bf5cb68f0d1392 : SUCCESS in 1h 45m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/35fc64ba916a4664aa707905aa83ef7a : SUCCESS in 1h 45m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/332c5a3ae5e143e88774cbb1fd47175f : SUCCESS in 36m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52120a80b0934e39887fe4e1b06d6510 : SUCCESS in 59m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91026d24a5e64e399bc2bfaf15336107 : SUCCESS in 1h 31m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/766e2d6d0f5b4d80bce6c23dc5b5782a : SUCCESS in 31m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8a6cfed8a124c0290f54345c70aaf03 : SUCCESS in 1h 30m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/77ae1c382f0c446dae5176a65a9c0d8a : SUCCESS in 1h 24m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc7a2e1d315b467cbe4b7df0e88d0244 : SUCCESS in 1h 45m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c773e128cf24139a872b09b4de48e22 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d316c58f1d9d9d2cd0bd3e9b7200ea0c999c7f51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:05:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4871f591b7bcc9f55140d1e36277ff5d440d13d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-22 16:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_07_2024_16_05_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ca7b21fadc3e65137053d53e56b1d27c30d3aaf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-22 16:14:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/924631/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/38a0574/ : RETRY_LIMIT in 3m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/924631/1/upstream-check/cinder-tempest-plugin-ontap-nfs/32a66c5/ : RETRY_LIMIT in 3m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a19cf2ba2a71f352af2551442456bea475edf676","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 16:29:40.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/924631/1/check/fujitsu-eternus-dx-iscsi/57b43e2 : SUCCESS in 2h 22m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/924631/1/check/fujitsu-eternus-dx-fc/6c578bf : SUCCESS in 2h 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"57b73efcf2ae18c15350b6e653dc6edf3aec01fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-22 16:57:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_Unity_FC/329be16/DellEMC_Unity_FC/2276 : SUCCESS in 1h 31m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_Unity_iSCSI/0aecd11/DellEMC_Unity_iSCSI/2283 : SUCCESS in 1h 47m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"20cf6d616ec65e36ed3d622e014c37b66363be9e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-22 17:05:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1795a8e3672d6bab707166e4a3248f3dab7779d4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 17:23:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerFlex_3.6/75d0f0d/DellEMC_PowerFlex_3.6/1764 : SUCCESS in 1h 56m 51s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerFlex_4.5/ab1c224/DellEMC_PowerFlex_4.5/643 : SUCCESS in 2h 02m 48s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/9043884/DellEMC_PowerFlex_4.x_NVMeTCP/601 : FAILURE in 13m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0c341206bb19254781537af5d4caefd97fb383e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-22 17:59:15.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5a1725c2786140c4a683b8aeace1f7a3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0e4a764dcb2c4f939ecdc24db9d833bd : POST_FAILURE in 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"46b8d94002cb3efe00413a3137cf66ffa7e917d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66eb00681d984466878c88c589c90489\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7574761be114206b1257c4d49423397 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94f50086b1a140f7944f9799e49ffa2a : SUCCESS in 12m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/462e4e5af673476c841f077bafe1945f : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d237030505f44ce4a6b31e5e0d3b4388 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/0d314a9752184166bcbf036c5e797634 : SUCCESS in 1h 02m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/567b9402ade94d7096c5c559536b56ad : SUCCESS in 1h 21m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dcbac2abee24404bf8df0470b2b2f06 : SUCCESS in 49m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0856482837d845baac2877fa3df95b41 : SUCCESS in 1h 15m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3249e32ed7804e2eb13e1611bf044ab8 : SUCCESS in 1h 45m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46e019e8f80249979b073dc26ae52f77 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9990e9bcba0d726f9112faf40d48c609cae1128f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:19:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ba3bf355efebc137593c2d7b58b2a8b9528c959","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 18:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3407511f1adf41c2ba79eb87efb01309\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8c3e78c33e41d3a229fddf5e03798f : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"01cacf157ace02fcfb707afe0243ef0e5d28e611","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-22 18:57:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerStore_iSCSI/44cf486/DellEMC_PowerStore_iSCSI/1531 : SUCCESS in 1h 17m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerStore_FC/14539f5/DellEMC_PowerStore_FC/1728 : SUCCESS in 1h 04m 41s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerStore_NFS/ded5af2/DellEMC_PowerStore_NFS/1226 : FAILURE in 1h 10m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerStore_NVMeTCP/d7e1b4b/DellEMC_PowerStore_NVMeTCP/1085 : SUCCESS in 2h 02m 44s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/31/924631/1/check/DellEMC_PowerStore_NVMeFC/5d7c497/DellEMC_PowerStore_NVMeFC/47 : FAILURE in 2h 52m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","revisions":{"834d933fa9fef6630d5df7ae278c6dad6dd82b80":{"kind":"REWORK","_number":1,"created":"2024-07-22 13:55:30.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/31/924631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/924631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/924631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/924631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/924631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/924631/1"}}},"commit":{"parents":[{"commit":"8a6ffb36c90379984836b3d04406277020b72445","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a6ffb36c90379984836b3d04406277020b72445"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-22 13:52:59.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-22 13:52:59.000000000","tz":-240},"subject":"Pass with mypy 1.11.0","message":"Pass with mypy 1.11.0\n\nInclude some temporary skips to pass with the\nnewly-released mypy 1.11.0.\n\nActual fixes to follow.\n\nChange-Id: Id3990df4fbe5914cc3b5e10ccf90aa9e5c1210a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
