)]}'
{"id":"openstack%2Fcinder~924660","triplet_id":"openstack%2Fcinder~master~I660ce14a663e406cb127db8da9a87fc336022c2e","project":"openstack/cinder","branch":"master","topic":"adjust-mypy-job","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-13 05:35:56.000000000","reason":"Someone else replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-09 15:51:36.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-02-09 15:51:36.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-11 14:40:12.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I660ce14a663e406cb127db8da9a87fc336022c2e","subject":"Enhance mypy job","status":"NEW","created":"2024-07-22 16:29:35.000000000","updated":"2026-02-14 22:31:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"81dd7f33499b2893003a5c5c5105be92e6f28e20","_number":924660,"virtual_id_number":924660,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-09 18:09:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 16:29:46.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-22 16:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 17:13:20.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-22 17:20:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 17:39:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-22 17:51:39.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-22 17:52:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 17:53:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-22 18:38:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-22 18:48:08.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-22 19:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 19:21:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-23 03:42:57.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-23 04:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-23 05:59:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-23 12:22:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-24 15:40:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-24 15:41:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-24 17:28:14.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-10-12 14:25:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-15 10:32:22.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-02-09 17:04:00.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-13 05:35:56.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"}],"messages":[{"id":"805636a1e8cb2cdb23051c9d23d15d943efb8472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-22 16:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ed19c15258cc4c03c47c51bf156c78bb6126c6e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 16:29:46.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/112/","accounts_in_message":[],"_revision_number":1},{"id":"484f8a10001e779629709f8472ec8a40faca5788","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-22 16:29:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/112/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"e8001b107d583e4c087e6313286546a4e730233b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:48:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf2f81f85554f039ed233335a885d83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67bae354b4dd4fb79f21d17e2b2aa06a : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1deee2fed8254415a9674f6e165552ea : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9800a1715629ad6445b81e2abb3147e4a4b69c0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-22 17:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924660 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a2575afbea0137e8e44ff5601188586856a09a23","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-22 17:20:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924660 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7c69cb953c35673ac0cc1b40bfbee62e19383453","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-22 17:39:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924660 : FAILURE in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"683f722ea35e8c6f509842126da1ce3a0c997209","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-22 17:51:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/773/.","accounts_in_message":[],"_revision_number":1},{"id":"2d63bcbcea2f8043fec7e336d940e887c79058c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-22 17:52:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6821783092ea422b8eead109b3aa46e4 : FAILURE in 1h 20m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1f6ced3d212e426eb7b44f583f63768d : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4bebfffdaa9cd4a05fa5d99cda5ea258aef6707d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-22 17:53:23.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/abd376722e4f452e91061358ee3caa52\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c89622e2682d44a5a9fee120ccff3398 : POST_FAILURE in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"93d8774390f3929131fefbf60a2cf4fec00e7edd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-22 18:38:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/924660/1/check/DellEMC_Unity_FC/1fca71b/DellEMC_Unity_FC/2277 : SUCCESS in 1h 27m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/924660/1/check/DellEMC_Unity_iSCSI/4dcb5c5/DellEMC_Unity_iSCSI/2284 : SUCCESS in 1h 40m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"58e6fd274025e0a44337c30f16a283b6a52b6bff","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-07-22 18:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/60/924660/1/check/fujitsu-eternus-dx-iscsi/f720819 : SUCCESS in 2h 17m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/60/924660/1/check/fujitsu-eternus-dx-fc/4026e98 : SUCCESS in 2h 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"02cb0bd2d31d83a93ca56a38390404d611f0cbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 19:04:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72419adfbd35403497ba524261e525cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281b6ad949f340628c073fd0f6f134e1 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/266d481ab12947e7860262733585f091 : SUCCESS in 10m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7870a2b079c845e0b7630368b705f798 : SUCCESS in 10m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f8c200362b240919db6e6f31dff3871 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951e4c2b9b4a43c3becfcbc493d27962 : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f49c64225bff498fbd213f1dee295dfa : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc8a9d77f656463d9531f069817c3e4f : SUCCESS in 1h 18m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ba7b060382e4f4fb8ebe2a20b94fc28 : SUCCESS in 58m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb343db760b342c3983768b8615e2c2b : SUCCESS in 15m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9dadc9ce441e421ea6a3d1b59ff96334 : FAILURE in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94be98423a71418189f08898c39c9dca : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96fe82f3d5cc45dc8593bd42456e8479 : SUCCESS in 6m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4ec582eabefa48abb08f671eb8074eb8 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6cc4f37c6e3a416ab8a2573e33967620 : FAILURE in 28m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d50579eb9747468aa366cd0fdce1216b : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5fcf165bc2fe4820b330fea09800a2ca : SUCCESS in 2h 16m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8544efe5d0454f6cb7d7e419a741e169 : FAILURE in 2h 04m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78154f00385e49adaff84fd0deae999d : SUCCESS in 1h 39m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c22bc83f336748d9b2be0a41c9dfed09 : SUCCESS in 2h 28m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/47dc0e7c98ce44a88bf31191c0d33318 : SUCCESS in 27m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/41c08d7050e34ca7ad1b9da6d394e02f : SUCCESS in 1h 01m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61482c83a3944c14803edfe76976fa0a : FAILURE in 1h 18m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e66e692fde14e21bf85012d56cd1830 : SUCCESS in 47m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bdffbccace194694b99e2ca3a5cb6fbe : SUCCESS in 1h 24m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e2c1b5d7e7d341a9be25e2dbd6c3f8b2 : SUCCESS in 1h 37m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56b372b42e5d4dafa95ab63dbf8dd222 : SUCCESS in 1h 29m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbe30f97eb04540be929c1c88f95402 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4ea6e676bbd944b130e6d0b51caae8114ba08d35","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 19:21:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/924660/1/check/DellEMC_PowerFlex_3.6/794c65c/DellEMC_PowerFlex_3.6/1765 : FAILURE in 1h 59m 09s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/924660/1/check/DellEMC_PowerFlex_4.5/009b47e/DellEMC_PowerFlex_4.5/644 : SUCCESS in 1h 57m 43s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/924660/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/6710e30/DellEMC_PowerFlex_4.x_NVMeTCP/602 : FAILURE in 12m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c28df422480b888fc60646c407cbcaf8a8b179b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-23 03:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/1/20240723084608/NEC-V-ISCSI : SUCCESS in 3h 55m 51s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/1/20240723084608/NEC-V-FC : SUCCESS in 3h 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e5fd2e97a8740afa52f5d4b4f7a60771ef59a49f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-23 04:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/924660/1/fc : SUCCESS in 0h 42m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/924660/1/iscsi : SUCCESS in 0h 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5b3e274e708ffc1b3830c48c28487176c7085ca8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 05:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/1/20240723091133/NEC-ISCSI : SUCCESS in 5h 19m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/1/20240723091133/NEC-FC : SUCCESS in 5h 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"70c778c1211e7c93e836cd421366c22a338bd3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-23 18:22:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_5314\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"_revision_number":2},{"id":"16f628f1032fca3235148ef615e149fbf3cb8a31","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 18:22:09.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/122/","accounts_in_message":[],"_revision_number":2},{"id":"08d86648ac4bd8c385a06b94f2c2dca52d53c135","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 18:22:11.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/122/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"c2baf5705069d3c9713605981f7a193ad8d3228d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 18:33:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/770c241fbc234d4ca5972e9afa0525d4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92ce1358cad543cea4baf72e88536bde : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b11b82f388624c6f8e0e5815cadb1122 : SUCCESS in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"817dacd16d9d81858856055cbeb9db927e3d6714","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 18:47:11.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6cacc698e2524e3da5003b2e93bb8d94\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d4402ec7704a40c99e35970ecb3c3819 : POST_FAILURE in 21m 33s","accounts_in_message":[],"_revision_number":2},{"id":"dbffe701df66ec5ab6dce82fb8c1859c92016271","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 18:56:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924660 : FAILURE in 32m 46s","accounts_in_message":[],"_revision_number":2},{"id":"67a42ee40f49b4fc025a13613699b3691cd4d897","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-23 19:07:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924660 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d62c8570afdf874ebbd0743c27f083da18cd2997","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 19:13:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924660 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":2},{"id":"169ce5d4202dd62815a77b3968d3b8f519ffa6c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 19:40:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5fc77312cd444a8789b504c9f0d84e8f : SUCCESS in 1h 16m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/30933e89c2fc436c8a7d3efe327fbc69 : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"abc404ae6e955336b21239f1a4dee1c269455029","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 19:46:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/782/.","accounts_in_message":[],"_revision_number":2},{"id":"0bcf8bb0deb232384791627ffc8910ff7a5c8692","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 20:37:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_3.6/38c5dc1/DellEMC_PowerFlex_3.6/1774 : SUCCESS in 2h 02m 22s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.5/9aac41b/DellEMC_PowerFlex_4.5/653 : SUCCESS in 2h 01m 29s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/6814392/DellEMC_PowerFlex_4.x_NVMeTCP/611 : FAILURE in 10m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"9cb593b1c65d8520af05e83bf16daf1d4d5bca88","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-23 20:37:37.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_FC/32c8b14/DellEMC_Unity_FC/2286 : SUCCESS in 1h 55m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_iSCSI/1079654/DellEMC_Unity_iSCSI/2293 : SUCCESS in 2h 04m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"22e3a9d3bf3b907c42b06fb7c781e0e6bd6763c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 20:52:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86feccaefc664cfe8faf30d64c2e59a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593c9ef71e0b4a65a64628ad4aa37c10 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbb974cef951459aa32ea0bb8a5bc5ec : SUCCESS in 13m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/febcb9bd58204488914c1d290c971a22 : FAILURE in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64bc357ed5be44f98327fb26dee1ff89 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69c21fde0e840ea9c9198b0374b2bbb : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/53db6ae6ddaa46af84313917f050d7c6 : SUCCESS in 51m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55c1a8df167a435e95305cc18ae3580a : SUCCESS in 1h 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e245a29706745a1b6d6074e707845a7 : SUCCESS in 1h 02m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e17b5280332f4ec083fc49997b62079f : SUCCESS in 17m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49f6dde68bed41378fecc0988687004b : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e17fab842cd4657a3a1b2fb9d4be83f : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3b69d7221ba41688f61627625b6d149 : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b177fc95c44c449490a4ea402f84859a : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05cdc11e156845aba0f9e4db21464d4a : FAILURE in 39m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0fba0ed82c5e4708b066181de74c404e : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8fe221f541140e592375f58df3d7c75 : SUCCESS in 1h 56m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11e2c476b7ee4e9398e1c246839fcd2d : FAILURE in 2h 22m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1d41d9af15a4b90bc95338234fb4e11 : SUCCESS in 1h 50m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e8b00ed6a7f640ff8bdfff5e0a28cc14 : SUCCESS in 1h 58m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9a1f48dbbb5647cf8a67383e1017cecd : SUCCESS in 30m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f46c16c703b478e9aedc3b10eae620c : SUCCESS in 1h 10m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7769588159944c8cb406f1ea36c8410e : SUCCESS in 1h 51m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/779cc1181f5547029a4b67954b473a55 : SUCCESS in 47m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a53f37f586c423eaf1847701d381c9e : SUCCESS in 1h 34m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/418ffecdecbd4d0d88fd6031a7d02a3c : POST_FAILURE in 2h 08m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/638edc84c16144f9b389803c573e6028 : FAILURE in 1h 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e155679f41a447ce96673b4da721c311 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7846c02f8e20353188c00398c9d56b2b6fbb8ad9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-24 12:55:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d6527d123865786eef645da087b906a8d7c7ebf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-24 12:55:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d3f69f71089c2bbf732b03b394b24c8148d4aef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:05:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76acb4c0a004241808db5567495709f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29b2b31e1019479b87430d7e0843a51e : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16370b2fe53b41b28f32a9f73125a5ff : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6508659a455424c39636e3e95ced7ef6e2e0783a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-24 13:17:34.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1034cb1b0d944a64b6cd4f40efdba266\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/968a4fb45d634f1e8f0f1925f1359d8f : POST_FAILURE in 20m 29s","accounts_in_message":[],"_revision_number":2},{"id":"af26b1625e9f1a23d43c19ebe5bf7b22893e4c45","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-24 14:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b78e83ce78854077b7ee4cead5a021bc : SUCCESS in 1h 19m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/acc37400a78e45a2ac83100a9f64dd15 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4f1db1cea73b9dcb9731f7c2daa541e28bc9453e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 15:31:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b0b01322a44ad3adaf5bc8ab4e4864\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/886e92cf0f1d4161b7d7b39d9bfd052f : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47bc0c56fb084df0b09069c58a704afa : SUCCESS in 10m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0fdf66767c42178dbbee3edbde43cd : SUCCESS in 10m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9de0d8ea3d4548fe8412f294e3533e9f : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a138122e80264ee7a77369aeff4197c3 : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/8f9aca2612bc4304aab2c436ed2f7a83 : SUCCESS in 54m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8393a014425c449c884c800e096da3e5 : SUCCESS in 1h 29m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/732c8b7012254a499937dbaae9750fbe : SUCCESS in 53m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77d0adda80274fc1b7cc6b3f34dd8e58 : SUCCESS in 17m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1985ae5b54b1483a8eda14190723b22d : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43e2b999dd1e4173b1de563b2311fa2d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c4f8bdbbadc4532825f50055d71bef8 : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/64b4c07b5bce40a58973f65dcfcedea6 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c48ec9eeca24532a6d47b128d0c0fd4 : FAILURE in 28m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bae4b5f28c6e41068e917680d38c24da : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cab05c6db663425e8235d837f1db8177 : SUCCESS in 1h 37m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b9f06819a8e3466cabb1d6df5dea1ea1 : FAILURE in 2h 29m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d538966d83b4d038fe3db764d0406c8 : SUCCESS in 1h 30m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ab7fc99a46784ea1a5de55048e423efb : SUCCESS in 2h 14m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/725e4b081e174fba97106fbcf3652717 : SUCCESS in 28m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/275a3275d48f408695872a9f6925365c : SUCCESS in 1h 12m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bb724a4591934d0780b05fbb158f1037 : SUCCESS in 1h 26m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c3e1197517254f87b8ca675c294ed750 : SUCCESS in 44m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d560883326c54df9bec3d8fb4a2874e7 : SUCCESS in 1h 21m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e07d82eaedab46ecb1e4807942e5c64e : SUCCESS in 1h 27m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9949b4d576a142179a48f957613c6905 : SUCCESS in 1h 30m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaf2c4126c774f2eb02de59e9be0369b : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":2},{"id":"edbe58a9ffe6d5fbb6e8c510235cbceba8f1feb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-24 15:40:17.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_07_2024_15_34_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"6f79d3a876001a78c4fe99f7e0d54e860ebaf726","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-24 15:41:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/924660/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/94b2a22/ : RETRY_LIMIT in 4m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/924660/2/upstream-check/cinder-tempest-plugin-ontap-nfs/c967325/ : RETRY_LIMIT in 8m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b1480563d9644e78aca06406a8dbb59470f752ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-24 16:22:38.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/2/20240724170010/NEC-V-ISCSI : SUCCESS in 3h 46m 40s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/2/20240724170010/NEC-V-FC : FAILURE in 8h 22m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9aa37978795e4ad209cf4c6c0eb4b0a199cc1dd8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-24 17:21:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_3.6/0a8b733/DellEMC_PowerFlex_3.6/1779 : SUCCESS in 1h 46m 58s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.5/f943549/DellEMC_PowerFlex_4.5/658 : SUCCESS in 1h 49m 16s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/a9787af/DellEMC_PowerFlex_4.x_NVMeTCP/616 : FAILURE in 13m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b42fcff0814edb23abac1da0ec1f241cd0d3968","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-24 17:26:53.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_FC/dfc7a5c/DellEMC_Unity_FC/2291 : SUCCESS in 1h 37m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_iSCSI/aebe000/DellEMC_Unity_iSCSI/2298 : SUCCESS in 1h 54m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"69fa91dcb992f39a9500fa792c5d39c7f9405968","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-24 17:28:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerStore_iSCSI/57d329c/DellEMC_PowerStore_iSCSI/1540 : SUCCESS in 1h 31m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerStore_FC/77f2eab/DellEMC_PowerStore_FC/1736 : SUCCESS in 1h 06m 38s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerStore_NFS/b750742/DellEMC_PowerStore_NFS/1234 : FAILURE in 1h 16m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerStore_NVMeTCP/ee15061/DellEMC_PowerStore_NVMeTCP/1093 : SUCCESS in 1h 56m 22s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerStore_NVMeFC/2de43ab/DellEMC_PowerStore_NVMeFC/57 : SUCCESS in 1h 11m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"62a01e33236cd8c0500cd3ff80ff587ce978b9dd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-25 04:05:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/2/20240725044329/NEC-FC : SUCCESS in 5h 42m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/924660/2/20240725044329/NEC-ISCSI : SUCCESS in 8h 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fb9aee9d2f4f72361e22038e0716f77bc2fba238","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 14:25:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f2e4a9261d97be3917b568ca67af915a988e76cf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-15 10:32:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4bbf9f275d80a5ad885fcebdce5cf0a8cccf98f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 10:52:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4476b1642884dfca6aa967b2791f8d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/261394fd149848e890b8564dce9851e3 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e65f3da1430c460a8c8494433bbf8e60 : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbe9edf9ffcab69ae90ba02726edfec3e1a02413","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-15 11:09:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/d60af69f58264dc3a19b30a7237602ab : RETRY_LIMIT in 4m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"df99eafeae6fc4ee3ac000c330ad25519194c5d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-15 12:00:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee21edb0ecd740b19ec37005d5a3f670 : SUCCESS in 1h 19m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7ed8dc4fda654af6ac01bd4281ed200e : FAILURE in 1h 25m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0aa95736ab7c8d2c5e874d771444a3aabfd52ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 12:49:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d56d76b83194f0f8f395098213f6193\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776248961cd8493d838fbb7ee66c0e02 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b662f66065a4de8a43963d021ef100c : SUCCESS in 12m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/096a968ee4a94875a3e62a766990eb62 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00bea920f1e45fe8046bc5c44af4772 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/51b3e2c47f2a4124b554f3a978d4a000 : SUCCESS in 56m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35d0dc48b26343cf9412e76aab328e6a : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88d28124cd84416e8d50f33b03aaa7b2 : SUCCESS in 1h 41m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/949f7d35b34840c3a3c3f9a60a3d2ca0 : SUCCESS in 59m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d33671824f94cb2ba4fd35c38496419 : SUCCESS in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6292ccc5d314fd5ac0e577526b5bb2e : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/253f24d09d48489b8a0957e1f9110e04 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07d515d0b7544c6b82f95eaf0b5da52e : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3430e7348b8940bbbcf462aadd4d4be8 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f38fa280ad64f75ab58b502cc12a6be : FAILURE in 32m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e87e1835902e437ebcbf24b26076a044 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b0208334a084870be30a777df137989 : SUCCESS in 1h 49m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ebbb3e3c71e4871845e44ac7f733116 : SUCCESS in 2h 14m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1450a6524f8e44d9b5ffb8b2f05adead : SUCCESS in 1h 49m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/53d1dd2005aa49f68c3a609391ea05f7 : SUCCESS in 1h 42m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eaf01de06efc4e2ca5d2e980610b26e5 : SUCCESS in 23m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b160794e51c74022943ce712b5433a1e : SUCCESS in 1h 13m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e45617fef1f4fd79378923d0d1707e3 : SUCCESS in 1h 18m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7f4b2dcc49944c428fe866318fca1cf0 : SUCCESS in 32m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/55ee38a8af8f4772bd4591d482c3b140 : SUCCESS in 1h 28m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/faeb93444eda4584aef49d5425cb0eba : SUCCESS in 1h 25m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d74e5b5113148259c5d93642c66a13a : SUCCESS in 1h 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc237fcb304647f1b54e9a3aa82465ee : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":2},{"id":"757846c253c5cfe2cee3001860a531ab05357aca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-15 13:06:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/797fb3d85f7a4e048736a793509a926f : RETRY_LIMIT in 4m 54s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/723aa0e527d94d4897b84a4a75c4d91b : RETRY_LIMIT in 5m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1a760af2970023fb054ef58bc9888387eb50a15","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-15 17:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"681caa9984c3040c81b4cef22d9bda8627e56c22","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-16 16:16:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_3.6/3c71cc6/DellEMC_PowerFlex_3.6/2436 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.x/4c25692/DellEMC_PowerFlex_4.x/1284 : FAILURE in 1h 47m 26s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/7e4e043/DellEMC_PowerFlex_4.x_NVMeTCP/1177 : FAILURE in 4h 58m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"157a898777c0f82edda3154f372eda7b7afe12ff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-17 01:00:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_FC/c8559dd/DellEMC_Unity_FC/2888 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/924660/2/check/DellEMC_Unity_iSCSI/a7fa19f/DellEMC_Unity_iSCSI/2855 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ceb57154ac8c44a868841d27882dec2e080aa32e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-09 15:51:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1bbcf235836e9c0c7a01f769cd0e88ea0155bec7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 16:11:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5854d3ac5534b5ba7097b1e97cc50c6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af0169a7c6c9437089e77a19737154b1 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f88eb58e1f44be88c70c2ba7af31e70 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f912dc2ca644d71ac3222143d8ebe0a : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5204aa9842f9599710d1c0f42572d9eecfcb38f3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-09 16:56:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924660 : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":3},{"id":"1dd9fd9d5a2f0370245f23c4feec126cd14936b8","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-09 17:04:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/29233c04023a4b59a4eb9898607e02d3 : FAILURE in 21m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ebc6a8accd0c9604d3b3edb8644b0e484221902f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 18:09:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f086e4cda61b45429b5de72699b9351b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/386c4e23979b43beb07a67f55e288d8e : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb05a97f5fbd49b2bca35a37656cd95a : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f993d377227465d802b2e1cf3a6ed22 : FAILURE in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3ce4a506994b7a9be2e90ee3cc5b9a : FAILURE in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/66b9b7bfa4974319ae31f2a567cdc6fe : FAILURE in 6m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e062bc1119294c3997dfca1a842485a4 : FAILURE in 5m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f64cee1f99544e839cdb54886cfb1042 : SUCCESS in 1h 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35aabd2a2fd7446e9f1fe1cfff619842 : FAILURE in 45m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd5b3e959e6144feacb97eb353f7c263 : FAILURE in 5m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f2d36ac15cf4d63a9f88533c4d57d39 : SUCCESS in 2m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4ca732038b5b46799b7fd7915beb610c : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a56258a43652430799df73c22d73d5bf : SUCCESS in 5m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f4a93e1ef5844350a2407c8fc0809a62 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d7a062925af84531bc21986b79f5d690 : FAILURE in 41m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d454e667ac00460bb158120bf0addf9c : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e68578b55ccf433f9964ca574e687bc7 : FAILURE in 1h 01m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f0bfbad478714c67b9756c82e1291eba : FAILURE in 2h 10m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8d63a3065314d5a988d14b3d0c9c14f : SUCCESS in 2h 12m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cfbe2360cc094f1a89ee68eaee3a3a59 : SUCCESS in 1h 54m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2b8c927942b1424b94229115c84e8272 : SUCCESS in 39m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9905bf0104534e77a51438e01c51af28 : FAILURE in 9m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8496c1c7366244a18e397bdc9b421797 : SUCCESS in 1h 34m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5ce20e02e124e7ab1eea8cde82dd539 : SUCCESS in 42m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f06928abe32044d2bcfaa6f856c6001a : SUCCESS in 2h 08m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/421eb69a09414d4ab3979db4d7529d4d : SUCCESS in 1h 28m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96685c82b4804a60a4f49d183a67aa30 : SUCCESS in 1h 37m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc7702f5640345119bcdc9b5a3353c9e : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e1214583ba34ba62a3977732ed17895242a07d05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-09 22:17:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eca60c20f341435b935c19ed48c5ca95 : FAILURE in 54m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5fa0cc1862c849a8bbf40f7bf6eba79b : FAILURE in 1h 05m 32s","accounts_in_message":[],"_revision_number":3},{"id":"30bb7f269484a7a0853cd603880def8aa72d3e6c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-11 14:40:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f85d63bf00731ab0673b9747cb3aa64356cc1282","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:35:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7dda2f2d03be48468bcd7120cb7dd7e4 : NODE_FAILURE Node request 200-0000623382 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"81dd7f33499b2893003a5c5c5105be92e6f28e20","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-14 22:31:44.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/924660/3/20260214212552/NEC-ISCSI : FAILURE in 9h 48m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/60/924660/3/20260214212552/NEC-FC : FAILURE in 10h 05m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"517fed2c228b35af2681c15629a8f2f1a4ca84fd","revisions":{"754c300cc9836d24b22cdebf27409816299d5aae":{"kind":"REWORK","_number":1,"created":"2024-07-22 16:29:35.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/60/924660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/924660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/924660/1"}}},"commit":{"parents":[{"commit":"8a6ffb36c90379984836b3d04406277020b72445","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a6ffb36c90379984836b3d04406277020b72445"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-22 16:25:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-22 16:25:58.000000000","tz":-240},"subject":"Enhance mypy job","message":"Enhance mypy job\n\nOut the mypy version info in a conspicuous place, and adapt the\ncode coverage job so that zuul will display the HTML mypy report\nas a job Artifact.\n\nChange-Id: I660ce14a663e406cb127db8da9a87fc336022c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/754c300cc9836d24b22cdebf27409816299d5aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/754c300cc9836d24b22cdebf27409816299d5aae"}]},"branch":"refs/heads/master"},"ab8a4aadabc6a469120b1c664e638ba3d873a739":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-23 18:22:00.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/60/924660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/924660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/924660/2"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-22 16:25:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-23 18:22:00.000000000","tz":0},"subject":"Enhance mypy job","message":"Enhance mypy job\n\nOut the mypy version info in a conspicuous place, and adapt the\ncode coverage job so that zuul will display the HTML mypy report\nas a job Artifact.\n\nChange-Id: I660ce14a663e406cb127db8da9a87fc336022c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab8a4aadabc6a469120b1c664e638ba3d873a739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab8a4aadabc6a469120b1c664e638ba3d873a739"}]},"branch":"refs/heads/master","description":"Rebase"},"517fed2c228b35af2681c15629a8f2f1a4ca84fd":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-02-09 15:51:36.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/60/924660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/924660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/924660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/924660/3"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-07-22 16:25:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-09 15:51:25.000000000","tz":-300},"subject":"Enhance mypy job","message":"Enhance mypy job\n\nOut the mypy version info in a conspicuous place, and adapt the\ncode coverage job so that zuul will display the HTML mypy report\nas a job Artifact.\n\nChange-Id: I660ce14a663e406cb127db8da9a87fc336022c2e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/517fed2c228b35af2681c15629a8f2f1a4ca84fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/517fed2c228b35af2681c15629a8f2f1a4ca84fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
