)]}'
{"id":"openstack%2Fcinder~924942","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I367df9b52f787403a54fb148713d122e9bb2a54f","project":"openstack/cinder","branch":"stable/2024.1","topic":"bug/1925809-stable/2024.1","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-14 16:54:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I367df9b52f787403a54fb148713d122e9bb2a54f","subject":"Fix: [Ceph] Backup Driver Python3 Encoding Issue","status":"MERGED","created":"2024-07-25 14:40:16.000000000","updated":"2024-08-14 16:55:59.000000000","submitted":"2024-08-14 16:54:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924942-bug/1925809-stable/2024.1","cherry_pick_of_change":920073,"cherry_pick_of_patch_set":1,"meta_rev_id":"910fc9b6086b6a0a38aaec5ab9b032ccfcfddd2a","_number":924942,"virtual_id_number":924942,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-14 16:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-07-26 13:59:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-14 14:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-14 14:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-25 14:40:16.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":36978,"name":"Wu","display_name":"Berlin","email":"1652124020@qq.com","username":"Ber1in.Wu"},"state":"CC"},{"updated":"2024-07-25 14:40:39.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-25 14:49:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-25 15:28:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-07-25 15:37:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-25 15:39:33.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-25 15:54:09.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-07-25 17:15:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-25 23:36:48.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-25 23:56:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-26 01:14:01.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-26 13:59:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-14 14:52:16.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"0f951b3b654d0f35b1e5e05eb0956377ca43d979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-25 14:40:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"75b93b90e396de79d3330983519e2ddd257f9054","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 14:40:39.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/145/","accounts_in_message":[],"_revision_number":1},{"id":"dabfdd0275bba4cdaddc0ea2a945380bba29e7d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 14:49:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/145/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"b704f2cd9505c93e26f14d9c18f1807fd1856eb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:49:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d64fc47b1d4a8e8a65d4c2d4ef714a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c341680402141318da3a66eac31faa4 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf176ac62a484d39990826472e581ea9 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b3137358ffc493d198e25708b25b92b1847bca4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-25 15:28:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924942 : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d86963cc7c46edf148765f2bf048c01755336217","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-25 15:37:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924942 : FAILURE in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5e0f7d4be2eccfc6235a4d42f9f25c0774896f4f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-25 15:39:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924942 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5ec20f4af873da8d3ef0bb0943dc353eb6b11996","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-25 15:54:09.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/966b3cbdd7344ee0a81d31ef04819e84\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/db16724647f34aacb40ed253512871af : POST_FAILURE in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e1f4174eff3f096b7922751fcb67d0bdd224cbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 17:15:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87a917b7d73b490fbd63675a547d454e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd9d158581646a190b387f1aa7574a8 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990dae9045a8448996abe2cd22ec7be9 : SUCCESS in 10m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d37f00c890524991bc3622043da7dcbd : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012b6ac29b964b1f8438fdaf60818675 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/13dabbd5f3a0416387039950f615aa61 : SUCCESS in 1h 05m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fd993c2682b843e0b9668fcd30a2e9e1 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f027a4b81df84677b7e24881a480e93d : SUCCESS in 1h 40m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c18cbebbf344a3bb28c725a4b7568d2 : SUCCESS in 1h 02m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/25c380c8e66841b5a21c104814145895 : SUCCESS in 21m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a4a28db5431a4b0d849cc8666af8222d : FAILURE in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3825160fe8a40278d183eec30430096 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3baea86fe86544aab848c53151acebff : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e11554b7b46b4ea99d3a769721868fe7 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/998cedf12aa54b9cb0fd09fe55506fb3 : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/572d613cf1b9409ab50b8be01a0f66b6 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/009c3dce0127440890a8eca7fa7e5eda : SUCCESS in 2h 04m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/555523f5c0c64d76ba2c8ea9337e95f3 : FAILURE in 2h 28m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1d4af4fbdaf48b2829242977a03c9ac : SUCCESS in 1h 44m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bb1ba94971c34688b7c078b625578b53 : SUCCESS in 1h 49m 26s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b024e2c8e9da477e8a4358d3fa94e9cf : SUCCESS in 1h 05m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b18e664d8494464da7a476fe42aecdaf : SUCCESS in 1h 44m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/03643e6bc7734623b7da67985d1ba0aa : SUCCESS in 34m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21b2fa46b49641d8ac4922284a2d9009 : SUCCESS in 1h 15m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9cb25be1c0d242adadb49a82d01a4f4c : SUCCESS in 1h 38m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e79f294804741ab9b059dba12708c1e : SUCCESS in 1h 41m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb48bcec0f6e4944b668ec5751f92747 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b17b9ee3cc9d81ce3fd3bcba1becdcb1306474a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-25 21:28:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f3e2658959dbed5336d989833568efde32f4d698","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 21:28:30.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/148/","accounts_in_message":[],"_revision_number":2},{"id":"96bd5c5b7e496f87fac1c9fdebc3f367d534be32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 21:37:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e146a50cf5bf4a719895cdaa6e44dfeb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76e7f9d1650947f39eed0866a2c9fe8d : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/25136ced804d48bbb28d4e2849c7ac33 : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"880d6022fd12162d542fdcdb41c6691cea36e40b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 21:38:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/148/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"bbf0cb9e0971d4e5ca3c9bbc78e85ce6dbfe45da","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-25 21:54:32.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3c9f8d3c0153467d84b39c093c043f85\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4369c2e009544255b47b45afcd8e8e2e : POST_FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a566afa8e97cb14e73383e7ef8a7a45aff293dea","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-25 22:14:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/924942 : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":2},{"id":"df492cbb9480682d0491000c3c690a667748d846","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-25 22:20:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/924942 : FAILURE in 33m 38s","accounts_in_message":[],"_revision_number":2},{"id":"604b242e59af5263a4a7fee0a6c5ecf0d1f8dc50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-25 22:32:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/924942 : FAILURE in 33m 19s","accounts_in_message":[],"_revision_number":2},{"id":"868ff1126e873be5cfd6363253d5a235a0c8df94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 23:31:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457cb01da50a444384dca0feb16aba3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a875b349444f457db7dc70ef8367f635 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cff2f73e46b4b00a2ede271624062fa : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5459e55eec26449da344573f7813f91b : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7a0214e67940fcbd5f1522d5b3de9e : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/dfa661cffd3048009f6f2acf2d1beada : SUCCESS in 1h 03m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5116cf4be4c3424c95fdcc42d56216d3 : SUCCESS in 50m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08e1f12bba0a4937aa9135ebc9598435 : SUCCESS in 1h 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6dd75892d854fc2b64665c9e00693c5 : SUCCESS in 1h 11m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/243f03351ed348199dc57f60f9d465d3 : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0b014eb545d44f298b5f7a047a04543e : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3dabf961ace4b8aac177ec4d54b53e1 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35fba8b32fa94a7ebc8e08b8c354aadd : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c93a1597b224bb0a69aa7077a96df56 : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3e128a84101f42278415f071046dd27b : SUCCESS in 44m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/664b40032f3e450c9d2b52bbaafdbc75 : SUCCESS in 4m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da9a28b8ee504d76ba394c4e6412d934 : SUCCESS in 1h 56m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a8f20f7c685e45fdad11244f327aab42 : SUCCESS in 1h 49m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b63c2ff942c24d428b338e9bd9b96ad4 : SUCCESS in 1h 34m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/233b71c47ca548bda6dc2d871c20b6a9 : SUCCESS in 1h 51m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d76d3def1dfd49f9acebd879072e2c76 : SUCCESS in 1h 14m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/37bb0b1181204dba8ecc183ffb1c9aae : SUCCESS in 1h 07m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5574c2b2f9d044bdaceac41a14ab8afd : SUCCESS in 32m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9caba59ac3d943529cfecd2358f1509e : SUCCESS in 1h 12m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/abdcdb0b3f1141d5aaade4b42d305f25 : SUCCESS in 1h 06m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6bc6442f0644b08b17fe82be163889c : SUCCESS in 1h 13m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ecf1805be7407a9e2c3cb1db81d689 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":2},{"id":"90f27869452760eb373470392d5d7dd9bd621527","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-25 23:36:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_07_2024_23_31_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"44a290bc05be65d8a308690dffc26690bc7335d6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-25 23:56:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/924942/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/7f50883/ : RETRY_LIMIT in 3m 18s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/924942/2/upstream-check/cinder-tempest-plugin-ontap-nfs/0118eee/ : RETRY_LIMIT in 18m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b5b5da4f2e0dc503ddc1c2ead2f40bce2437056","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-26 01:14:01.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/794/.","accounts_in_message":[],"_revision_number":2},{"id":"7bc1ad077478b3c8f8a21ece468329cc2a4184da","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-26 13:59:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a42e1c1c4ac90353db0dc69ff5198ce1b1df672","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-08-14 14:52:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b8bd9912f0b0c70115c9ba1d06915cfab4c89c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 14:52:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6816862db22a2c689ad8d807ceae3940bdcd2daa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:54:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc9e44b0cd064ac58c88caaaf683a018\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e300188dec84cfba1f9a439401661da : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2e88df6151e4dd1b133e2faceb424b1 : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b71a283e0284a78a5489c80f419d744 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c44077ecef14fc297ae8b08bd1fd2d2 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/5e69faa5111245d2aaef46d02ea463bc : SUCCESS in 1h 06m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/99d8827195a24336af5f09a1b9d02aa3 : SUCCESS in 55m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7cfb6aad0e0e48c88388e05e0a03462f : SUCCESS in 1h 24m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31e64ce8e6af4e80b29ab6eab748c942 : SUCCESS in 58m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a9364b87c35459f8a460271f8dd52b5 : SUCCESS in 1h 23m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc20a44a1b31417ea167d6243678477b : SUCCESS in 1h 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50eb0cfac02c4ac99ee2d1f40ea088e0 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":2},{"id":"57d6d8523195d9661b124d779d3ca6e4fbf51a38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:54:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"910fc9b6086b6a0a38aaec5ab9b032ccfcfddd2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:55:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c63d38227b492cbd9140974a9a2361\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c5dab4f3f0460ea40c6e20a99404da : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da46f4d775e02b2a36eac8a8fb139728f173e3ab","revisions":{"9d4161e70f77173d9eba1f9ee6301e23b179db97":{"kind":"REWORK","_number":1,"created":"2024-07-25 14:40:16.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/42/924942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/924942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/924942/1"}}},"commit":{"parents":[{"commit":"6565c1fd2b58047c12b5a159a3e7836ad0f36637","subject":"Merge \"HPE 3par: getWsApiVersion now requires login\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6565c1fd2b58047c12b5a159a3e7836ad0f36637"}]}],"author":{"name":"bolin.wu","email":"1652124020@qq.com","date":"2024-05-21 05:47:33.000000000","tz":480},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-25 14:40:16.000000000","tz":0},"subject":"Fix: [Ceph] Backup Driver Python3 Encoding Issue","message":"Fix: [Ceph] Backup Driver Python3 Encoding Issue\n\nUsing python3 the Ceph backup driver faces encoding issues.\nIn order to solve this we should start using bytearray for\ncreating a zero-filled instance with a given length.\n\nCloses-Bug: #1925809\nChange-Id: I367df9b52f787403a54fb148713d122e9bb2a54f\n(cherry picked from commit c13370cdec32ae6b74f3653981519af567be40de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d4161e70f77173d9eba1f9ee6301e23b179db97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d4161e70f77173d9eba1f9ee6301e23b179db97"}]},"branch":"refs/heads/stable/2024.1"},"da46f4d775e02b2a36eac8a8fb139728f173e3ab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-25 21:28:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/42/924942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/924942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/924942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/924942/2"}}},"commit":{"parents":[{"commit":"3843e3ebcfed13ede5c5507a7111dd1f7dbd67a1","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3843e3ebcfed13ede5c5507a7111dd1f7dbd67a1"}]}],"author":{"name":"bolin.wu","email":"1652124020@qq.com","date":"2024-05-21 05:47:33.000000000","tz":480},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-07-25 21:28:23.000000000","tz":0},"subject":"Fix: [Ceph] Backup Driver Python3 Encoding Issue","message":"Fix: [Ceph] Backup Driver Python3 Encoding Issue\n\nUsing python3 the Ceph backup driver faces encoding issues.\nIn order to solve this we should start using bytearray for\ncreating a zero-filled instance with a given length.\n\nCloses-Bug: #1925809\nChange-Id: I367df9b52f787403a54fb148713d122e9bb2a54f\n(cherry picked from commit c13370cdec32ae6b74f3653981519af567be40de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da46f4d775e02b2a36eac8a8fb139728f173e3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da46f4d775e02b2a36eac8a8fb139728f173e3ab"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
