)]}'
{"id":"openstack%2Fcinder~925526","triplet_id":"openstack%2Fcinder~master~I9cbcf1f108ddd93f33c9b929144ed715100a0567","project":"openstack/cinder","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-07 14:18:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cbcf1f108ddd93f33c9b929144ed715100a0567","subject":"Keep new RBAC disable by default","status":"MERGED","created":"2024-08-01 19:09:25.000000000","updated":"2024-08-07 14:19:42.000000000","submitted":"2024-08-07 14:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925526-secure-rbac","meta_rev_id":"4a9c9182be7a0778eaf5022098009106049ee32c","_number":925526,"virtual_id_number":925526,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 14:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-02 19:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-08-07 12:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-08-07 12:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-08-02 19:00:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 19:09:35.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-01 19:10:28.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-01 19:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 19:49:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-08-01 20:01:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 20:09:59.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-01 20:56:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-01 21:31:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-01 21:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 21:48:18.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-01 21:52:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-01 23:42:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 23:43:02.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-02 07:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-08-02 19:00:20.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-03 03:56:30.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-04 05:50:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-07 12:40:03.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"09ef9f65f5817ad9985c16f8713570c3177b12bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318d45f0fe1af6590dbfd0e6b362ba04140f5ce1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-01 19:09:35.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/208/","accounts_in_message":[],"_revision_number":1},{"id":"3de0ed5d1375f651132aeaa2fb0b4cc1e8995d62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-01 19:09:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/208/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"42b22f63ca61b2f1b9a19bae2038806407a9b30c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-01 19:10:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"5215730c05bff55466d381078611a379125f2c09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 19:33:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f32f4b01abe4288bdb31008270a36f5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcbc1efbc4db48f4ba6a470162192515 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bd2e59fd54a438c98d00bbf7a7a4930 : SUCCESS in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb7bbb1708bdfeb0a6808e248e85d38fc598d105","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:33:40.000000000","message":"Topic srbac removed","accounts_in_message":[],"_revision_number":1},{"id":"06feea890ba3e7b654526d278c3138ce2b206093","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:33:43.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"094f5aea4f26ee370338bd3d654bb7a104f07e77","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-01 19:49:50.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2eef6f5f37f5432eadcb96fe7d4e839f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/24acf2c828874300b14ae5fdb34f9a51 : POST_FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e4e0e65cc300f193f1f5c310ecbe0ce1c166f9f5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-01 20:01:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/925526 : SUCCESS in 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"292d1c0f678a027a56140482fd0861fb8fb5d74c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-01 20:09:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/925526 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a4b36dba5422568f2200f429eb188b3375a3b65e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-01 20:56:50.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/833/.","accounts_in_message":[],"_revision_number":1},{"id":"fe90d1fc8051e6d63f89cfbad6417e0e9a6820b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-01 21:31:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_Unity_FC/4402d74/DellEMC_Unity_FC/2342 : SUCCESS in 1h 47m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_Unity_iSCSI/0bf6f15/DellEMC_Unity_iSCSI/2347 : SUCCESS in 1h 50m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"67ab6f22489e86a0534e3a87422f9ac9abf3985d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 21:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb55f737d8b245cc94e7ae42a5fa45d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ddbc0dc025402ea3fc8d6a3c15463c : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c05ceff93e74741976d59a45f47c3da : SUCCESS in 10m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfc4940764a045e4bde8a3acfea6e6b8 : SUCCESS in 14m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f8adbb2d15541cabd4ac9653b4bcb5e : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2245faa78345559a8a687b92924ad2 : SUCCESS in 14m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3597e20122754e9d9c504bd491a74b04 : SUCCESS in 3m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b6dd9d8e1564487d815e47596e3bdc44 : SUCCESS in 53m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8eb03303edc644efbc6e6fb98829652d : SUCCESS in 1h 25m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edebacb9846e45b0bf1ad6dc30bcea3b : SUCCESS in 54m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95bef933d5e84d4f88e265bccf844265 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2116db5cf52c4bfa8bf6e933b6e375e6 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfc34623455140809c43bc6f31bbedba : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c4258f328984faf82b5c186a5c8ebcc : SUCCESS in 5m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e8083f19f161480eb87c939bc89395e6 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a0715deb9ef484e9ec44ab42d6ec1ab : FAILURE in 31m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a4ca765fa1d40d29883d3ac7c46d5b7 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d3d169b2d6f4b2798bff8258de3a2c8 : SUCCESS in 1h 45m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e80c0882060e49158800cb631c8c1cd3 : FAILURE in 2h 27m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/abecb0b744fe46dcaf824bde07841e69 : SUCCESS in 1h 31m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0ffaa079d34c45ec9af265ea4d814563 : FAILURE in 8m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f52abc52840b47029a533a223725ad7a : SUCCESS in 35m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aac08b357291422698535d66edcc8187 : SUCCESS in 1h 16m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd0a58e12f284663bccd8399fadf9754 : SUCCESS in 1h 13m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5208d3f5c1141e6b4e68dd99fb1ece5 : SUCCESS in 36m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ff0b8cfa4ad4f6cb7a38850fe7887c5 : SUCCESS in 1h 28m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1ac9a191a643498bb342570160854570 : FAILURE in 8m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39f0c3cf708a4e6d999bc8ad44bb5b3d : SUCCESS in 1h 11m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87a5504e9b164c8e8751c1742da08b74 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6deaeddc1bcceb0010657abebfcedbda96b1e94e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-01 21:48:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_08_2024_21_42_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"8c88d64d1a9bd1dcecf568576aa4181f1968894f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-01 21:52:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/925526/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/47a06b9/ : RETRY_LIMIT in 4m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/925526/1/upstream-check/cinder-tempest-plugin-ontap-nfs/9a21ca4/ : RETRY_LIMIT in 8m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc9a1523218900ad41267e42913b523666cab3e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-01 23:42:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerFlex_3.6/1eeb567/DellEMC_PowerFlex_3.6/1829 : SUCCESS in 1h 48m 10s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerFlex_4.5/e38e2c8/DellEMC_PowerFlex_4.5/708 : SUCCESS in 2h 11m 06s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/87ab7b9/DellEMC_PowerFlex_4.x_NVMeTCP/672 : FAILURE in 13m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0f3420b344b51ed5b099574bb57527db2709d6fc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-01 23:43:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerStore_iSCSI/63f8341/DellEMC_PowerStore_iSCSI/1561 : SUCCESS in 1h 22m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerStore_FC/1740773/DellEMC_PowerStore_FC/1757 : SUCCESS in 1h 11m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerStore_NFS/51838ea/DellEMC_PowerStore_NFS/1255 : FAILURE in 1h 18m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerStore_NVMeTCP/4765959/DellEMC_PowerStore_NVMeTCP/1114 : SUCCESS in 2h 00m 12s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_PowerStore_NVMeFC/c627a9d/DellEMC_PowerStore_NVMeFC/78 : SUCCESS in 1h 16m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"0399befc0b8886aad2e92c0deee04c6ef470deb3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-02 03:05:35.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_Unity_FC/0e1ff25/DellEMC_Unity_FC/2345 : SUCCESS in 1h 41m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/925526/1/check/DellEMC_Unity_iSCSI/51ca78d/DellEMC_Unity_iSCSI/2350 : SUCCESS in 1h 44m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5c2e7b95ae99afc4adc19769e4af5671402dc764","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-02 07:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/925526/1/fc : SUCCESS in 0h 40m 39s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/925526/1/iscsi : SUCCESS in 0h 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"aec27f4cd43c3143107e39c8f04737bc10abd7da","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-02 19:00:20.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27b99c47b9b4cf77a955b9c5b56e497d502e6c9a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-03 03:56:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/925526/1/20240803090212/NEC-V-ISCSI : SUCCESS in 3h 27m 28s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/925526/1/20240803090212/NEC-V-FC : SUCCESS in 3h 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"50fd67e5fb0ff938e7916c60882ffd22195d8dd2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-04 05:50:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/925526/1/20240804062352/NEC-FC : SUCCESS in 5h 49m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/925526/1/20240804062352/NEC-ISCSI : SUCCESS in 8h 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0706c8064ab029e427b7d47f1c865c666f05596d","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-07 12:40:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44c636b85c0ce90e4e511a4800d7b6a924b09c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:40:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ecd21dbc3b03325f96dccca34340f9dbe583582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:18:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f0d6505cbb46df934594d9523181d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0133bcb0ed82406c9a1ee26fc2fd4082 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc1461678ae4d13b369042ecef9bee0 : SUCCESS in 10m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00397cd90fd54fee849db4c10dea515f : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a056b5d2344fe8a464811712ef9e31 : SUCCESS in 11m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d543573976d343f6b53863f99231cde6 : SUCCESS in 3m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/af29466d28b54c2f9ea6e1653d9909ae : SUCCESS in 45m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fcf586bd71234ed68ab76fedc276e10a : SUCCESS in 1h 25m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/861d393452d24fee96a610b3c7e6000c : SUCCESS in 57m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38c9cf31de8a4c69b7a88e8c5e19f28d : SUCCESS in 1h 25m 31s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/787bf547c9da460ba88edcc0fd69134f : SUCCESS in 1h 37m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66df7cad15ac43119f8e1524b822c902 : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7cff29fd18c48f3aa3b6b10771d14d059dbcf4a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:18:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a9c9182be7a0778eaf5022098009106049ee32c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:19:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20cedd2d44b94cf18a976a8f74cc3419\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15071612e57c42fcbbe082cdc7b1b83a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45330f91d5f93ea7f053f260a2ada9396139fb1f","revisions":{"45330f91d5f93ea7f053f260a2ada9396139fb1f":{"kind":"REWORK","_number":1,"created":"2024-08-01 19:09:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/925526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/925526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/925526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/925526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/925526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/925526/1"}}},"commit":{"parents":[{"commit":"7382bea3a7e380879f53f31a47360e5059c141f1","subject":"Merge \"[Pure Storage] Driver API version upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7382bea3a7e380879f53f31a47360e5059c141f1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:04:29.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:09:22.000000000","tz":-420},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil Cinder is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, cinder continue with the\nold RBAC as default.\n\nref: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: I9cbcf1f108ddd93f33c9b929144ed715100a0567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45330f91d5f93ea7f053f260a2ada9396139fb1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45330f91d5f93ea7f053f260a2ada9396139fb1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
