)]}'
{"id":"openstack%2Fcinder~925722","triplet_id":"openstack%2Fcinder~master~I4e38d592abf1d36dedda97a17710746b4037c779","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-08-27 06:37:23.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2024-08-28 22:41:21.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-08-28 14:51:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9535\u003e replied on the change","reason_account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-16 05:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I4e38d592abf1d36dedda97a17710746b4037c779","subject":"Added support to Active/Active mode in ISCSI/FC drivers [NetApp ONTAP]","status":"MERGED","created":"2024-08-05 17:12:47.000000000","updated":"2024-08-30 06:42:47.000000000","submitted":"2024-08-28 22:41:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"925722","meta_rev_id":"26eff0829a02517e32e3a20c87fdb3c84b3f0965","_number":925722,"virtual_id_number":925722,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 22:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"recommended":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"all":[{"value":1,"date":"2024-08-27 06:37:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-08-28 14:10:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":2,"date":"2024-08-28 18:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 14:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":1,"date":"2024-08-28 18:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-05 17:12:56.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-05 17:13:51.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-05 17:56:47.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-08-05 18:12:30.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-05 18:17:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-05 18:43:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-05 18:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-05 18:55:07.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-05 19:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-05 19:35:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-05 19:42:09.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-05 19:50:07.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-05 20:55:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-05 21:01:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-05 21:17:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-06 05:28:14.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-06 05:28:14.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2024-08-06 05:28:14.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2024-08-06 05:28:14.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-08-06 06:09:27.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2024-08-06 06:09:27.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2024-08-06 06:11:49.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-08-06 07:44:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-06 09:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-08-07 05:01:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-07 09:27:58.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-08-21 04:17:48.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2024-08-27 06:37:23.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"}],"messages":[{"id":"645df00f9a37f0235e5c6e7bd1120a68f51a7b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-05 17:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b8009bc836496a43ea6efefd029ee76212068b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 17:12:56.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/220/","accounts_in_message":[],"_revision_number":1},{"id":"d37f806366e38b8ec58ec7bd0bfe7930f80bf891","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-05 17:12:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/220/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"8bdc3fef24d4d552cdda8f7a3a11b4c705880cda","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-05 17:13:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1b6a6331441985a0faf33479d7863171a61b2e7c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-05 17:56:47.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7a0613d24a6f40f19c2cfdfaf7cb7906\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1a61939656ab47da96f204550704ae09 : POST_FAILURE in 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"381646d401fff9e12830729d66ab25055296f2d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-05 18:12:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/925722 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":1},{"id":"df254b9206398780694a11e24624f824082b7845","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-05 18:17:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/925722 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"953786572edf780a01df59ae9f7acb6d2e104844","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-05 18:43:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4b1751c5ca4244c692c44cd832b1271a : FAILURE in 1h 28m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f0c643aef2484bc9b3ece3e8e03a1bfd : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8da432ac4f367327832a9638e3411a0a2be94f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 18:45:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f0aa285cdb4a7295354d9ff358e0d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29e077f5dc5648d9a52b0866bbb9d616 : SUCCESS in 34m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a196f92234ef450da05e33f8ec905fa4 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f55ddd2b6cbda38f784f96b33f4c8b80a540363","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-05 18:55:07.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/846/.","accounts_in_message":[],"_revision_number":1},{"id":"b14ad9041d93019643b0d4927b81c792ef3fd7e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 19:29:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e050125bd7954aaaa5beaf65cc082168\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f6a6e9429d94dd8b4e67ad46a56f7b5 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7530c16d3b0549ecbae5ad0758c54bab : SUCCESS in 11m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fad1f92bed8643ac9c6e3835c5b71f1d : SUCCESS in 9m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7051f0b24bd423888dfd853f7ed4e2a : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7f0453a7e14823ba4c12c5583e7987 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/fb4adc8287c942808a99771a56084dfb : SUCCESS in 51m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bbc9d64b007247a1b8c36ed2407f03d0 : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f659db4107b24cbba6d78cbe98c9f5ad : SUCCESS in 1h 02m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c759e8d15374153986ce40377984ec2 : SUCCESS in 17m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5e51fcacd2f4d6985f9146e8e41791f : SUCCESS in 18m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11d985dc83e0473f8c3d7280f41952e6 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1bcb620f044c4f17864eba3164d57d79 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4293d6c4f7464e4d89f49273f41bb083 : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3aec5d50b76f45cf91a96207a2aef9dd : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a3ceb523cb61462bab1fada05fcbeb7f : FAILURE in 38m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/848ae4b81f8449289b2b77c4f2207b7e : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af233fcf31f24bdaafee547f2babface : SUCCESS in 1h 58m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1d5d511e98384d8692a5878e1c1ddd28 : SUCCESS in 2h 14m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb4b34a94789472084e271d6b5222567 : SUCCESS in 2h 00m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2a09e22c00ba4f7fa8b07df2e6626dbf : SUCCESS in 1h 55m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d6a71a03e7074c13931495f0e48a90fa : SUCCESS in 23m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cde263d157ad43008624b01c03981dd5 : SUCCESS in 1h 18m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6267712137ac49649e483a86f1cca085 : SUCCESS in 1h 18m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e8872f05799f4ad6a187007c60cf97a2 : SUCCESS in 29m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db9663917451469abf1914480e7db1fc : SUCCESS in 1h 08m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2f0524453417491b85f8d0df22a90e8a : POST_FAILURE in 2h 09m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2eef7a1a1b25477192282a07216c8c71 : SUCCESS in 1h 11m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee14e36f3c8a4eaf9ee655578191698b : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"555329a75bf8514a946ccf62850c9518e5389863","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-05 19:35:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_08_2024_19_29_30\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"663acaad5ccf926dc5909e42a7d250806c9d33e7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-05 19:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/22/925722/1/check/fujitsu-eternus-dx-iscsi/873d28c : SUCCESS in 2h 28m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/22/925722/1/check/fujitsu-eternus-dx-fc/9dfbfd1 : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"092db0caeb2f6a3650e29a0ae58334be343dc9dd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-05 19:50:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/1440814/ : RETRY_LIMIT in 8m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/1/upstream-check/cinder-tempest-plugin-ontap-nfs/fb0569a/ : RETRY_LIMIT in 3m 40s\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/24043bc/ : RETRY_LIMIT in 4m 17s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/9b9c32f/ : RETRY_LIMIT in 3m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"52881942c52580641236771e68b4066e001fa1e5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-05 20:55:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerFlex_3.6/75380b9/DellEMC_PowerFlex_3.6/1845 : SUCCESS in 1h 47m 00s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerFlex_4.5/b87f500/DellEMC_PowerFlex_4.5/724 : SUCCESS in 1h 44m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/ce40882/DellEMC_PowerFlex_4.x_NVMeTCP/689 : FAILURE in 12m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"98a4e0ec9ada1590d7065c53942d546e346a4f29","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-05 21:01:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_Unity_FC/d42446a/DellEMC_Unity_FC/2359 : SUCCESS in 1h 33m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_Unity_iSCSI/c976c46/DellEMC_Unity_iSCSI/2364 : SUCCESS in 1h 46m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a588dccb78ecd694cef47cf031117f7ca7989b6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-05 21:17:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerStore_iSCSI/08e5fa5/DellEMC_PowerStore_iSCSI/1571 : SUCCESS in 1h 27m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerStore_FC/ed2b5c4/DellEMC_PowerStore_FC/1766 : SUCCESS in 1h 12m 03s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerStore_NFS/c5cd8bc/DellEMC_PowerStore_NFS/1264 : FAILURE in 1h 18m 13s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerStore_NVMeTCP/672c6e2/DellEMC_PowerStore_NVMeTCP/1123 : SUCCESS in 1h 48m 21s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/22/925722/1/check/DellEMC_PowerStore_NVMeFC/8500ecb/DellEMC_PowerStore_NVMeFC/87 : SUCCESS in 1h 03m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"13a4d440640bc86de4d3380d19af2f15140c70ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-06 06:08:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3ff2b210b7754bbf4513af9e1c7d9b23654b5880","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-06 06:08:32.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/222/","accounts_in_message":[],"_revision_number":2},{"id":"36e07728e8571d6f77b334fa0a15629ae53d4bc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-06 06:08:34.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/222/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"bb5ea1d147897223cf6516a6e40af0904eb5390c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-06 06:09:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"eb7aea110ea8cc33f9c0d2655e41f9bc28ad571a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 06:36:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00b6c7a2ae4b4bf4ab38d8f85e223f19\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1c1929d6aef4fac844da5c86ba0cdbe : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f0970e63b63489cb43db8d5fb74d4e1 : FAILURE in 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54d1e07dd2853fe6a791187be4a5485b5244e3d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-08-06 06:38:11.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d9daa1b8641a4aa8975d66b49d3ec8cd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/393ca39a727946d3bdf7aa2040d712a4 : POST_FAILURE in 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7c3b6be1eb1c5dfa39033f40f1ab7ad993e4915e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-06 07:22:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/925722 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3ca22c88d0cf4c65db3ba1d52f74ee71cf0582d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-06 07:29:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/925722 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":2},{"id":"17c5cf6b4455196e247e7536ad62cf0948f58981","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-06 07:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/1/20240806122153/NEC-V-FC : SUCCESS in 3h 41m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/1/20240806122153/NEC-V-ISCSI : SUCCESS in 4h 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2dd495dc028db0bfd75869e0d1ce003b08831f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:13:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d59206e6bf4ec280112f1bfcba80a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176e3b13450340f7a63da3bb56a397e2 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60ac4f1bb8404b51a28636bd3410dc8e : SUCCESS in 8m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48ebe0d6907a4e178aa3032a1ca3d976 : SUCCESS in 8m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4fb3fd9830147efbe1ac27c47168150 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2eae4914a14be7b34a1e35ed24928f : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/0d3ff3b6586946ada324eeebeb8f725a : SUCCESS in 55m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce6fe78649df4649837c95af1957b9bb : SUCCESS in 1h 16m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66085e3685ea4f8bbcdbff91a38ea6fe : SUCCESS in 1h 07m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1374c6334b949459807149062824cb8 : SUCCESS in 21m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4569a1e5b2704a04a7a21442985e539b : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a0d4c0ebb6f44756a6934dd395ab7cc1 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b49816880034d90925d5d0ffdb29d8c : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a5852dcb626477b97a6a3d328019a72 : SUCCESS in 5m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bc0a70a2b7f845cc92c609cb5805071a : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5fce25fdf6f40748042598e3a012fbd : FAILURE in 39m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72bbadbc6b0a4e7ca42a7ec192b4626d : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fea68dc32ee463bac1429eab5f4fcd3 : SUCCESS in 1h 41m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81e1bbd5c2f84959814f6aa925010009 : SUCCESS in 2h 04m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7bbaf553bbcc45f78309fdd4aac406e9 : SUCCESS in 1h 50m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c986a502edfd42f986bfbdb2cfa0f712 : SUCCESS in 1h 40m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/020d0fe1410042c7879551009c479ba2 : SUCCESS in 21m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d0e8029d866145b4982d97ffb5974552 : SUCCESS in 1h 02m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b1352852b7447d89fa25e8ab7e7ad13 : SUCCESS in 1h 22m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7aee1959fa0748c09c49524148316e16 : SUCCESS in 24m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c582c3b00b14952b11ab9f8d474834a : SUCCESS in 1h 23m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c1e430964dff456bb85c5b3cb4fc023f : SUCCESS in 1h 40m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/945d9cd85b6f434091524670d3271fe4 : SUCCESS in 1h 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c86b6dc08a6448594d50e349dc24978 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a6e9452677e0620811122bce00dd2c453777d0c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-06 08:18:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_08_2024_08_13_30\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"eef731f9f8ecafbe8d86190d4ddcd4147ca66b32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-06 08:27:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/251352274c3f4c4b976645b9cbefb596 : SUCCESS in 1h 23m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3f8a909d49a346dba35a379f4c8172ec : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f868e4597f12af52837efa886c9f30fdedd4581d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-06 08:33:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/efa7bd9/ : RETRY_LIMIT in 4m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-nfs/cc8a5ac/ : RETRY_LIMIT in 9m 16s\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/d64a488/ : RETRY_LIMIT in 8m 04s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/f42243a/ : RETRY_LIMIT in 3m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e83fba1ee4f274d332cdc00b0f5ae58a3778adb2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-06 08:58:32.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/848/.","accounts_in_message":[],"_revision_number":2},{"id":"08baae796026b81e03cb94aa3f6a18af7b398241","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-06 09:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/925722/2/fc : SUCCESS in 0h 35m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/925722/2/iscsi : SUCCESS in 0h 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"497afac1ffe2cb9d379fa7ef6f3f80eb03d3cc86","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-06 09:47:55.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_Unity_FC/996f066/DellEMC_Unity_FC/2361 : SUCCESS in 1h 28m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_Unity_iSCSI/6ffd926/DellEMC_Unity_iSCSI/2366 : SUCCESS in 1h 42m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0c510b75b6d92fa47985cf267773ac67d0234270","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-06 10:04:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerStore_iSCSI/28aad59/DellEMC_PowerStore_iSCSI/1572 : SUCCESS in 1h 22m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerStore_FC/a32f38e/DellEMC_PowerStore_FC/1767 : SUCCESS in 1h 06m 03s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerStore_NFS/f169c21/DellEMC_PowerStore_NFS/1265 : FAILURE in 1h 08m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerStore_NVMeTCP/4afcbb7/DellEMC_PowerStore_NVMeTCP/1124 : SUCCESS in 1h 51m 13s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerStore_NVMeFC/ddcea8a/DellEMC_PowerStore_NVMeFC/88 : SUCCESS in 1h 06m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"b16764b33a76a18e086035b6d837ca473d6e3a60","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-06 10:40:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerFlex_3.6/12c0286/DellEMC_PowerFlex_3.6/1847 : FAILURE in 2h 57m 00s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerFlex_4.5/2f0a996/DellEMC_PowerFlex_4.5/726 : SUCCESS in 1h 58m 45s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/22/925722/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/a4c9acd/DellEMC_PowerFlex_4.x_NVMeTCP/691 : FAILURE in 14m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e759d8db25e6c3bea22ea0f6515eb2ce9ae39daf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-06 19:42:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/2/20240807010005/NEC-V-FC : SUCCESS in 3h 39m 26s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/2/20240807010005/NEC-V-ISCSI : SUCCESS in 3h 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"dbe44edbc939b88f448f4e16d6bc862f27d7d840","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-07 05:01:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/2/20240807081558/NEC-ISCSI : SUCCESS in 5h 15m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/925722/2/20240807081558/NEC-FC : SUCCESS in 5h 45m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3169a6463d30503d1f51ada5021022c354fb764b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-08-07 09:27:58.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9504c6930d2e4726eb0779957015fad937de23bc","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-14 15:10:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"99d310a619bfd2d19ea85860d4751e61d8e76e90","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-08-16 05:03:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a53e830119ab73fb21605cf1ff82ee1477406afa","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-19 05:14:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e74a661be7af103fa89a5efc028a698bade33c2","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-19 12:16:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fadcb41053a3d32ebb1c32ab6864ba363845905","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-20 12:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20f834c7ccda3379ab9f4cca1b8e7b301139b853","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-20 12:54:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd7f1c9382c61290e72f5e4154bf28dda9b92f25","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-08-23 06:49:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82b06153d6dfdf26c595c33f0940d3e1cc6b3a83","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-23 07:41:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-nfs/c13be97/ : SUCCESS in 45m 32s\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/da1d94a/ : RETRY_LIMIT in 9m 27s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/925722/2/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/f84382d/ : RETRY_LIMIT in 14m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7495c298d23aa93e062e366945d8188c9d317645","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-08-27 06:37:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"777ef9e9370bb95969637ae2b1ca669b96407924","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-08-28 14:10:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cc93a7c3cde373f7ba98a16850999b4a6ed593f1","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-28 14:51:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a5ad5de22e3132ef3aaa8852117b945b62ed9f8","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-28 18:48:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9973b05f439b63167ca49777353bca8a38c37165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 18:49:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5796c68c3d6cfb268551085f81aaf54b51a1668d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:41:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d2700e56ab64de09c4b93dce9d8b2c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8637efa2fe2048d9ae15b772fdb3fafe : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1937321c283b409ab38744826c628002 : SUCCESS in 12m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66aadf63d10144858d362caebe47fbe3 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d8fdac90e54cb5ae4313c81b0a6b7a : SUCCESS in 14m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/70b9486ec7e4408ca2d2a2d91476c64e : SUCCESS in 52m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/054341e8c6e34f7d9009da147e752bb4 : SUCCESS in 1h 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b286ee1de72488b8fe1428b16df2eb1 : SUCCESS in 1h 00m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3944a352224742a8967861b696378611 : SUCCESS in 16m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1d8bd9b3ee0400a876574939ca82bb9 : SUCCESS in 1h 09m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29381362f9d6401f8f7c8a4c228654cd : SUCCESS in 1h 48m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c56a50859539426e88a3567b06f56497 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":2},{"id":"174f8a85b6c634b9c5b736956f973cf243da947e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:41:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c168c10b8e24d70fb1d0e62a8089b78d32accffa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 22:42:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da06305b73dc4408bed5ae739d091682\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c420a377928844e1bec961658aa61f44 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aadf9d488569417299bab7ee2a3f3968 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"26eff0829a02517e32e3a20c87fdb3c84b3f0965","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2024-08-30 06:42:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"053b2fe63cdbddaccb8002b91d11065e9ce22af4","revisions":{"59fcf67067e645e19e02e93c8a98d5888a1659a1":{"kind":"REWORK","_number":1,"created":"2024-08-05 17:12:47.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/22/925722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/925722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/925722/1"}}},"commit":{"parents":[{"commit":"1e6d2d785c133caf869011b062766c394d2cffe7","subject":"Merge \"Fujitsu Driver: Add support for revert to snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e6d2d785c133caf869011b062766c394d2cffe7"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2024-08-05 17:10:51.000000000","tz":330},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2024-08-05 17:10:51.000000000","tz":330},"subject":"Added support to Active/Active mode in ISCSI driver [NetApp ONTAP]","message":"Added support to Active/Active mode in ISCSI driver [NetApp ONTAP]\n\nThis patch enables Active/Active support in NetApp ISCSI driver.\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I4e38d592abf1d36dedda97a17710746b4037c779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59fcf67067e645e19e02e93c8a98d5888a1659a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59fcf67067e645e19e02e93c8a98d5888a1659a1"}]},"branch":"refs/heads/master"},"053b2fe63cdbddaccb8002b91d11065e9ce22af4":{"kind":"REWORK","_number":2,"created":"2024-08-06 06:08:25.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/22/925722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/925722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/925722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/925722/2"}}},"commit":{"parents":[{"commit":"1e6d2d785c133caf869011b062766c394d2cffe7","subject":"Merge \"Fujitsu Driver: Add support for revert to snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e6d2d785c133caf869011b062766c394d2cffe7"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2024-08-05 17:10:51.000000000","tz":330},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2024-08-06 06:07:05.000000000","tz":330},"subject":"Added support to Active/Active mode in ISCSI/FC drivers [NetApp ONTAP]","message":"Added support to Active/Active mode in ISCSI/FC drivers [NetApp ONTAP]\n\nThis patch enables Active/Active support in NetApp ISCSI/FC drivers.\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I4e38d592abf1d36dedda97a17710746b4037c779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/053b2fe63cdbddaccb8002b91d11065e9ce22af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/053b2fe63cdbddaccb8002b91d11065e9ce22af4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
