)]}'
{"id":"openstack%2Fcinder~926298","triplet_id":"openstack%2Fcinder~master~Iefa3ecc8e68a832f88d7f63ba2734299587d908d","project":"openstack/cinder","branch":"master","topic":"LUKS-image-encryption","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-11-18 14:51:22.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-11-18 14:51:22.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-02-17 13:58:56.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-09-02 09:15:09.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-02-25 21:40:42.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-09-02 09:15:09.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}}},"hashtags":[],"change_id":"Iefa3ecc8e68a832f88d7f63ba2734299587d908d","subject":"Adopt image encryption standardization","status":"NEW","created":"2024-08-14 15:59:14.000000000","updated":"2026-09-02 09:15:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":30,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"40729a82c4ee4626bda2c5f817774e640fd606d8","_number":926298,"virtual_id_number":926298,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-10 15:12:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 15:59:21.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-14 16:34:21.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 16:39:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-14 16:59:26.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-14 17:14:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 17:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 17:58:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-14 18:04:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 18:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 18:59:55.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2024-08-14 21:09:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-15 14:09:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-16 11:55:54.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-20 15:10:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-22 12:00:44.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-08-22 16:59:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-27 16:37:31.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-08 10:21:54.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"CC"},{"updated":"2024-11-08 10:21:54.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-11-08 10:21:54.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-12-16 19:36:24.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-02-17 16:08:15.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-02-21 19:19:32.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-02-25 22:39:04.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-25 22:39:22.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-25 22:39:56.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-02-25 22:40:16.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-02-25 22:43:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-02-26 14:59:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-04 14:14:10.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-05-21 18:40:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-22 01:42:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-27 15:26:09.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-13 23:50:35.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-20 15:40:19.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-20 15:45:17.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-21 01:57:58.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-04-20 07:06:43.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-06-10 15:12:24.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"}],"messages":[{"id":"5e75f137e7d54474bcafebf8a8249197c26c8291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:59:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a954c890c1059ae53aa296da8e4fffd684d8c23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-14 15:59:21.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/299/","accounts_in_message":[],"_revision_number":1},{"id":"534cfb7f3db92c5b9a3920d934ca6b41dbeb328f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-14 15:59:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/299/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/1/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"22db766a7106c96a9dab9e44aff6bc6b249fa27e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:59:40.000000000","message":"Topic set to LUKS-image-encryption","accounts_in_message":[],"_revision_number":1},{"id":"c37df6aabbc0b8ef607cc933c99822eb556ecf0a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-14 16:34:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/98/926298/1/ : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":1},{"id":"e36e550f6a68c24ee852a5619c9aa2f3589ad34c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-14 16:39:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/1/check/DellEMC_Unity_FC/a2b90e4/DellEMC_Unity_FC/2419 : FAILURE in 18m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/1/check/DellEMC_Unity_iSCSI/1a15415/DellEMC_Unity_iSCSI/2422 : FAILURE in 26m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"50d160523514b7e88b202f9d10b86ad9a6c737dc","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-14 16:59:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926298 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"faa7a5a15b001b4eed3d0e803b7aac61d4804ca6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-14 17:14:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/1/check/DellEMC_PowerFlex_3.6/9df0d0c/DellEMC_PowerFlex_3.6/1901 : FAILURE in 30m 44s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/98/926298/1/check/DellEMC_PowerFlex_4.5/c385c24/DellEMC_PowerFlex_4.5/778 : FAILURE in 44m 14s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/cd627e3/DellEMC_PowerFlex_4.x_NVMeTCP/751 : FAILURE in 13m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"24eed9b1a3bf378b33f61e5be2a726ed613f35d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 17:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d9fe38352343e880f4deafc181dc28\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3dc4a1b59974489a86fd6291436d79b : FAILURE in 32m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3432d049568c4c80a335f514be993b81 : FAILURE in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e0a950d2c9fbc5cdbe3594f2b1643b190e21a9d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-14 17:58:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"61011040042a3a81a3b02ab8262615f889893915","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-14 18:04:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : FAILURE in 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"0c778c7fd084a4293a6ea0222564a1ed97d1a32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:27:28.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eff1b33c7f334e12b052a0fb79664dc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a636a99a6d4a72b0f32b0ecf7bbfcc : FAILURE in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8016fb29b93b4c0686a8cdc0b9c38205 : FAILURE in 10m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06f9d9f4079b4f7883ed080681288cdb : FAILURE in 13m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/469a3e9acd3444269be89e4379438dfa : FAILURE in 16m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c139f466124aa1b950026748362e99 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b3560c5e407943bfad02e2ab19e3fa03 : SUCCESS in 54m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef63deb816624bc98a5bc7459d95a72e : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/143cada09ae94f85b60393bff2a43603 : SUCCESS in 55m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/395ef051b4084e879999e5a2b9a1a2a0 : FAILURE in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/39320bf043754ec4b9c2fd2a9c5bcdbe : FAILURE in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d7af03975554ff38d71b515d859296e : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5584004310c4f3a84c1926634e7ef8c : SUCCESS in 6m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c8c6fa5b35a4138825743f40f518ac0 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/84dd4b8cfac04c0f930005d53a136956 : FAILURE in 21m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1720aaaa79446c68748926d583d0ede : SUCCESS in 7m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83e712a7099647c68178d5ce9b65a5bf : FAILURE in 1h 40m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d853d4fcb51d4301bd1b08c43216242c : FAILURE in 2h 25m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/713861dd95c94da3a9bda33b608b769c : FAILURE in 1h 35m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d1be01ce61fe4f378ee3fead8bbe0d3a : FAILURE in 2h 04m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/879264ef7d45448c980b1fb5044736cb : SUCCESS in 38m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8db1f4b71c8147a5ac574322f44b9697 : SUCCESS in 1h 09m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e21d829aaff14e2ea6d2c9f38f665f0f : FAILURE in 1h 39m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b99a1abef6b426ebc992ba6c8272e7f : FAILURE in 49m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34aee28c624340bbbf46eb83a37a1eea : SUCCESS in 1h 22m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/36373f03c89a4561a0ddea5b0a202b15 : SUCCESS in 1h 53m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e24b33fd40b34905a889dbfb669ed211 : SUCCESS in 1h 33m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e30ff0a53ce47d99c3becac0c4d0616 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0eb19eeef7c9265966b27d1e9a89e9015b5bc92b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-14 18:59:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f86accb19e6e90315b86f7b6050ad4e143a1192","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 20:19:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a210592860ca4dc08f635bd0b0af2c00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36ad80068b2c40b3a0ce31d51507bdb9 : FAILURE in 22m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4347e9b4614040939daeab85079ed123 : FAILURE in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81920a5f03498bdd7e81823ad196056ae323fadf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 21:07:28.000000000","message":"Patch Set 1:\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ead320912e584f74a298d8a9de73d429\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b78bef0cf4c4935907053459ad06ce2 : FAILURE in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95a5059f77344eb2afdaed42afdcd8cd : FAILURE in 11m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1710f88cbd5444fa226ab1c18157a5b : FAILURE in 10m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f58dfd9a70bf49e3868fb6b0a437a040 : FAILURE in 15m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d9eeb3846842498d0e220843b295ce : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a9343ab078db43dbad39ca41c0e0276c : SUCCESS in 54m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8cb1958f55540a6a3e26f5deb0a9b65 : SUCCESS in 1h 53m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdacd515dbcc4eca8b4ca431bc07e2f0 : SUCCESS in 1h 02m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1878c8fd30ed495db48b788bb86b26aa : FAILURE in 13m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/900e096483a84d10ac1b5cb981ea2829 : FAILURE in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b16207956ae463d92709b4ec9061550 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a738c9fdd2084bc4aa4eb50d94f7d75a : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2eb8eae3e2b0444fb630ceb419cc8448 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6bea986116ba42d889faa973892ee621 : FAILURE in 36m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e121444a40d468296155f638e0d40ca : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/868e580cf27f4fac8a40fefb3942dae7 : FAILURE in 1h 40m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ca8afac2f594d8a846cc0c043832db7 : FAILURE in 1h 58m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/778bd1b876134bacb98cc071c6bd9aeb : FAILURE in 1h 23m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dfec3f19c4c340d98bfaf76ea8251f5d : FAILURE in 1h 36m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf6b1fd52a76404291218dcbea049309 : SUCCESS in 27m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/57e3d2423d7b45099f414b483acba580 : SUCCESS in 57m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bbea43d06b7241f2972375f93842fff3 : FAILURE in 1h 19m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/219f0e50d0d94a61b90b9408068c65d1 : FAILURE in 33m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c109c83b9c247cc94a4313f8cba274b : SUCCESS in 1h 11m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/14e6dd4fd2744f21ba9478aba356bee5 : POST_FAILURE in 2h 06m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f499f77c024429096852311a1990a7d : SUCCESS in 1h 26m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a2bd907fee4ffcb06f56ca76f31c3a : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3d656a43191f83a912663aedc5031085812da34b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-14 21:09:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b7d510c7995a43a8a93e9aa9b48e5164 : SUCCESS in 1h 12m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/de41ed98fa394fd6935e8e9924a1262b : FAILURE in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"625492e87f00d640ad81fad45f7d9c169f773580","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-15 14:09:17.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/895/.","accounts_in_message":[],"_revision_number":1},{"id":"69979aaab38dc3d38e6a2eb68ad49d5fe326c06d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-16 11:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/98/926298/1/20240816162613/NEC-V-FC : SUCCESS in 4h 18m 05s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/98/926298/1/20240816162613/NEC-V-ISCSI : SUCCESS in 4h 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d6e3bde1dd563756da9b38b5f945296397fa7bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-20 15:06:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"08933ee5bd3f1a7ae7cb29332ae04c6b2e53f155","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-20 15:06:42.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/331/","accounts_in_message":[],"_revision_number":2},{"id":"e754b9dff69b549a1a94bed4b622f78723a6cb57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-20 15:06:44.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/331/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/2/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"694e316ff960157ca306e7ff8439ca1ffdd5baff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-20 15:10:34.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/os-brick/+/926293 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"302aa4408bf4524dea0275b7c7e5e4bff5004f24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 15:10:36.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926293 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"b24e644f8849f887e2a45e13ba732fa18eb075a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 15:10:39.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926293 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e368327649db6544ce6ee717d876cbabe79a24b4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-20 15:10:42.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a93761bc0e339f473a12e28662cedb4c2b7a58c8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-20 15:10:44.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"40696b751d76518c8a0fc26599e78ebffb4d13fc","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-20 15:10:44.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b25c8103072aafd3effd0464370dc47d73a38e4c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 16:37:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/344/.","accounts_in_message":[],"_revision_number":2},{"id":"ffd93125215127fba17ca9d161fe9bfb4dae469f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-20 17:53:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/10f358bb78094c459f04c3b7c91e8395 : SUCCESS in 1h 09m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4c79bbc65cee4619b0d9fb6ad353df40 : FAILURE in 1h 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"24a784984c8f78baa8e9fb1487e6f5f8c6f069ae","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-08-20 17:55:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/98/926298/2/ : POST_FAILURE in 1h 27m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c09658d1640f2e7c403707df1ddf283826c1ba13","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-20 19:01:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/2/check/DellEMC_Unity_FC/c16f569/DellEMC_Unity_FC/2458 : FAILURE in 2h 09m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/2/check/DellEMC_Unity_iSCSI/23009fd/DellEMC_Unity_iSCSI/2462 : FAILURE in 1h 53m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4bff4786a6ac001fe2f246eaa52a31023d3ca125","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 19:04:57.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/929/.","accounts_in_message":[],"_revision_number":2},{"id":"2b0038bf791c6bf2e845769dd27484016ec85d82","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 22:20:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/2/check/DellEMC_PowerFlex_3.6/a296b9c/DellEMC_PowerFlex_3.6/1934 : SUCCESS in 1h 58m 52s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/98/926298/2/check/DellEMC_PowerFlex_4.x/00db4a1/DellEMC_PowerFlex_4.x/815 : SUCCESS in 2h 23m 10s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/e972fe2/DellEMC_PowerFlex_4.x_NVMeTCP/789 : FAILURE in 2h 56m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"fffd38efe21e0d9bf0c3ee23c8482dde83948478","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-21 12:47:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4e7fe5a3494fb3a8415fafb4c732e0ca1c7eeb2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-21 14:59:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f7b8e97e941d4b26a073ea3922503cd6 : FAILURE in 1h 14m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/89045d7827734005949e7da01bb6799a : FAILURE in 1h 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"57336ccc17b3461f0c622e4ea4daa048341f05d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:32:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce9777e085b4a6c9172a8a8cb00caf4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79895e96b2fc4d7fb8ffb8d4a5c6a61c : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a4f5df6413b4e1cab8e66f582295d14 : FAILURE in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3720f768743d4b0c8aa7b94eb88d772c : FAILURE in 11m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a647e0b7876453793609ba1fe5beb4c : FAILURE in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea261c8eda34a42926eef68dbdde59d : SUCCESS in 15m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/aba591cc004746eb87aa22418fa529dc : SUCCESS in 1h 31m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bf7a73547804bd7a1e88a58035afe9e : POST_FAILURE in 1h 36m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4842cb51ada346298ae779ec13b948bf : SUCCESS in 1h 00m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52444e6dfa5d495bb2fcce9bfc4777c1 : FAILURE in 14m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0a8e1f930fff4cc5adb19734943c7fc9 : FAILURE in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e32ae39c97a745a894417d302a8d6298 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1145df821625437fa1a2155b7b64a0b5 : SUCCESS in 7m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/573e4982feeb4bd49462ba73b2f71673 : SUCCESS in 8m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c2962e0690649b89ef6be4cf8788ca4 : FAILURE in 33m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1cec8c1c59e94a6aae238e5ff97e857d : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb9af6fb55b144b8b893bb8cbedaa621 : FAILURE in 1h 49m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc0bfde859224971900237424775baac : FAILURE in 2h 28m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df58b6af99de4e518f46a8530dd02baa : FAILURE in 2h 00m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d695f914c8a14e54a009310aae6a5938 : FAILURE in 2h 18m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9a9608d3b99441819c0e33df08dffb79 : SUCCESS in 39m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b25b85aa99741c19aa1b06e472a19c9 : SUCCESS in 1h 10m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/430472039e6a4be6b20ff15174b48c50 : FAILURE in 1h 30m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/15b3b3cd401a4b6a8f39cb4f5835c8fa : FAILURE in 47m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e0138c94b8594c68be39d3e250e34504 : SUCCESS in 1h 22m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f7f3b4deef374152ad903a097662fa6a : SUCCESS in 1h 38m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00daf0fb8ae747af95aa3f9e7b054f53 : SUCCESS in 1h 26m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a1adc2b018c4f0eaf119eb1d5818856 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b4da48a841a98b2ec788872559dac05f28087652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 23:08:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497a209912ad4232a96ad7a7cf4f5ac6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c117f144231648da800592f9f6015995 : FAILURE in 46m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3971b77dfffe439989387652e0e1ffb8 : FAILURE in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c33c0521fb97fdc55ad9b87c8bb86d269d114f95","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-22 00:16:35.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/2/20240822013526/NEC-V-ISCSI : SUCCESS in 3h 45m 32s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/2/20240822013526/NEC-V-FC : FAILURE in 7h 41m 06s","accounts_in_message":[],"_revision_number":2},{"id":"54da3c53bf3d7761fbf979ecf7fbdc23ba9c0580","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-22 12:00:44.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/2/fc : SUCCESS in 0h 46m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/2/iscsi : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f810db830ed3387b71116416bed0796119561a0a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 16:59:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/2/20240822172357/NEC-ISCSI : SUCCESS in 5h 28m 00s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/2/20240822172357/NEC-FC : SUCCESS in 8h 35m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d34782e295c1125accf554e44c5196736ad52a5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-27 15:30:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b394ca836841d9b6686c941bf54d72d0db89f9f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-27 15:30:27.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/405/","accounts_in_message":[],"_revision_number":3},{"id":"0eeff003cab4c26bd54c7439f9c0f19b81428ec3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-27 15:35:05.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"d32c8b27bc061e8bfaf472a3b927d54a80c96f10","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-27 15:46:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e8482bc67f139d94b5eff69280478b1107e4a9c8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-08-27 16:37:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/98/926298/3/check/fujitsu-eternus-dx-iscsi/b91aeec : FAILURE in 1h 02m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/98/926298/3/check/fujitsu-eternus-dx-fc/bf68849 : FAILURE in 1h 01m 06s","accounts_in_message":[],"_revision_number":3},{"id":"84e7903f03246dc797e80dace04944e67dfdea54","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-27 17:04:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c84478982f884d229867916f1262e74b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e284a0a85904beab1bd62996cbdde4a : SUCCESS in 34m 44s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a2240ef665a4a63a615b3224eb7f273 : SUCCESS in 44m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f3ab1f974f740a6b971ffbeff6f4359 : FAILURE in 1h 33m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbd7bcec551c451caed92cc308030698 : SUCCESS in 37m 11s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"3c0fc875c873a9dc8fc86f8210a8795a56129c7e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-27 18:21:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 10m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e044790ef363e9ab0b0f5f231950d3fb31468b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:32:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69caa2ed4cee435ea5af8b1105e98400\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858713df29864248b57d138876a63d7e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f20472ddc6954c1a97a8a70f4175c3a3 : SUCCESS in 11m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/717c6e704f2a46cd93c640bcbf68df96 : SUCCESS in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bc276c231df4970a322e220d44ae601 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943f9d02f984478499b6648f66f18a99 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ab7eb34363e942f8ad1b893654433826 : SUCCESS in 57m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54a0d7da5eb2486eb0edabed676db370 : SUCCESS in 1h 36m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd0abcd248354a5a9f41cad88d20794b : SUCCESS in 57m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92ba7e3795be49ec88dedcbfbb9d96ab : SUCCESS in 21m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9ac561f402ce449cb6098cc27260d1e4 : FAILURE in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bcca70a6825a465cbf56b3ec54ca8a61 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28933bff129b4dde89f7f43257bb2cc4 : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b8f1d333d9f4e21aba73fbfc948e3f7 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb4899aa4b014135ad59d5093325f7c6 : FAILURE in 34m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4a394537ecb4dab86543ec540f22449 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/edf20f1c270b4797b8fb9daa7f0cb993 : SUCCESS in 1h 42m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5fa9a06acf884af99c31c6154ba7c9bf : FAILURE in 2h 25m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/582a928c72bf43d996f63978209eb318 : FAILURE in 1h 58m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0379fee511594dbc9a4e73f44940df49 : SUCCESS in 1h 52m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/168cf96e189d4b488b88213de038ab4e : SUCCESS in 39m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb24b8fd1755415ea5da8e84b4b80c1f : SUCCESS in 1h 12m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/023305e2787d46a7b2c6abe20d454780 : SUCCESS in 1h 25m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a5791e4aa90e43068c3034094ab27eb3 : SUCCESS in 39m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/449ef8dff9af413594d720e119b54a48 : SUCCESS in 1h 26m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/003862294f8349b59bed7f55b1ef1b25 : POST_FAILURE in 2h 42m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4cd4685162b460db377935443ba730f : SUCCESS in 1h 25m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a21c6b917da4666835becc25819f362 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c83ac8952fc23018e4ac113c41a3d0686f4d9e09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-27 19:32:20.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/414/.","accounts_in_message":[],"_revision_number":3},{"id":"ae15d9d74f812cc7d89f1678dd09d4034ee9c5c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:51:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f06073fea9545f9be8347460b3a1b27\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40068c6858604c419dcf4f9249611bef : SUCCESS in 22m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8eee59f8494b4fd391110e9a24c19343 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d238cc55b12fec51aa61dfa1e6992808dd3f20a4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-27 21:50:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f1ea6012716c4276ac340463fb5d3da4 : SUCCESS in 1h 09m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7139ab92f9c94cb19fa7b7110c9f4e33 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":3},{"id":"952b03a712bbe44382ef83d25ce1541d435fe7e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-27 23:09:29.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1001/.","accounts_in_message":[],"_revision_number":3},{"id":"cb255f05cb0ed3242762f64ec4f30ca7cede0f2e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-28 06:37:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/3/check/DellEMC_Unity_FC/16b7042/DellEMC_Unity_FC/2517 : SUCCESS in 1h 42m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/3/check/DellEMC_Unity_iSCSI/bc4062f/DellEMC_Unity_iSCSI/2521 : SUCCESS in 1h 59m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"eec792d0b0048c234e7b06a8d4aeea15aff2936b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-28 08:47:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1d64b4f65052c40cecce21a98faa9625c0e723f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-28 08:47:34.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/411/","accounts_in_message":[],"_revision_number":4},{"id":"a992dfbace43a5cfbb87981b00eab742e483c236","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-28 08:52:30.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"d32959a0f25dae67afed6d8c35244a517132ba14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-28 09:05:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 00s","accounts_in_message":[],"_revision_number":4},{"id":"4492a56a57f42660d60c92e659237e1c1cb66b54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 09:16:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aa439f8d22449f5a23259cea6e03aaa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/662f3dc77f524261b0846c18be6c6e79 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e00afb07067a47b588d2cfc9cd3acabd : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e8414410dce63a17b43c46779e10a7c2d4b0c22","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-28 10:19:45.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/904171d98ea54ae59fead76667e8b3d2\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c3cb87ab97a412c9de88913d1559020 : SUCCESS in 43m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f15463a64744e458fae0a8a38ff8a60 : FAILURE in 38m 37s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c312bc094db2411db0ba7c5805ac10ef : FAILURE in 1h 31m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18e26a98e1d44e6983f0d8ae019f2567 : SUCCESS in 38m 27s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"ae03064cc9769a0d6c213de51c8a1b2ff127b96a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-28 10:47:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":4},{"id":"4c3d4f23fb342703a0f7f54d9c4d7ff966e067f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-28 10:51:40.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/420/.","accounts_in_message":[],"_revision_number":4},{"id":"b97768b1d6a7ee22d042885a04523a299ce3a7bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-28 11:29:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f01dd6d7d0cf4cfeacd8aff766851c40 : FAILURE in 1h 15m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d93658f26ffd4f4ca2be209aca580848 : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":4},{"id":"95858258a8fc6d7d72afc10f1f0a7eac101e82b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 12:11:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d63380be95da4a6687bccc09aaf4c438\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5cdf88515d4b219da2ba904d096572 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5461f3275a948faaa0358c17f543ce7 : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb15b3685c0440868ccd85835bf95ca4 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/847451a9fbfc4f458b11f085cd0eef96 : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b26df49d4e4537b3fb7fecdf586da9 : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3a67b220ba0a456898578b8b5c526a21 : SUCCESS in 58m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a62e3c7e84fe4db4ab729b0cbcd9ea08 : FAILURE in 1h 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfee2f853e35434692183a57d70e5c36 : SUCCESS in 1h 11m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76481a6e9c92409e97c36a9e91ff2834 : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/595a2e755ea14aa89dd41ab5f817411e : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed9210b6747e4b5b9f2f0c6254532c97 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ced1341d5c14110986562cf47caa805 : SUCCESS in 5m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/24eb82714bd14edba9f1697a7ee1cc73 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb51018025ff4a7ca816876ad3bf4ff4 : FAILURE in 28m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f54076b41ff145bfb59585306300b96b : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2342af10f5a14074a116b623e5dee850 : SUCCESS in 1h 48m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/88020119f6d7423291954f5f18dd28c6 : FAILURE in 2h 33m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/70a152813d7e4a6d994ea8c4df228cc5 : SUCCESS in 1h 42m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1f3b14e2e09345aaa065629f1fde15a3 : SUCCESS in 2h 06m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c093c55ba55c419e95442aab0aa22d38 : SUCCESS in 31m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6dc415f0844a48b289297368354c1e55 : SUCCESS in 1h 07m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40098dcace9d4489accf91eabd16f6f9 : SUCCESS in 1h 29m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9d52075dcc84b76bfee94f017bb6910 : SUCCESS in 37m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6061a3f0cc5c4122b8d91dad76ef5d66 : SUCCESS in 1h 24m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/69f0cb203ffa4ce28f44c412603cac27 : SUCCESS in 1h 25m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44f7a4ae63b04f53ae324811c98b8328 : SUCCESS in 1h 35m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d4af51cadf44046bae2f4b403b19847 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":4},{"id":"a21f08a9712da8cfe263b8283c34f1d494d7e3d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-28 13:00:05.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1007/.","accounts_in_message":[],"_revision_number":4},{"id":"8ae652c30bea598f4ecc9af345bcfde16a04f280","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-28 13:06:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/4/check/DellEMC_Unity_FC/bb0a2e5/DellEMC_Unity_FC/2520 : SUCCESS in 1h 50m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/4/check/DellEMC_Unity_iSCSI/e3ee86b/DellEMC_Unity_iSCSI/2525 : SUCCESS in 1h 57m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"694a5c16b1a506b86c0c1b845c8862a39a6d5e70","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-29 10:28:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/4/20240829153437/NEC-V-FC : SUCCESS in 3h 50m 40s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/4/20240829153437/NEC-V-ISCSI : SUCCESS in 3h 54m 12s","accounts_in_message":[],"_revision_number":4},{"id":"527074b9a43eebeb8e7f7ebe07693e1f53d6a468","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-29 14:18:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/4/check/DellEMC_PowerFlex_3.6/9c0e295/DellEMC_PowerFlex_3.6/1983 : SUCCESS in 2h 25m 08s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/98/926298/4/check/DellEMC_PowerFlex_4.x/e2b16fb/DellEMC_PowerFlex_4.x/864 : SUCCESS in 1h 51m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/6a0a52a/DellEMC_PowerFlex_4.x_NVMeTCP/836 : FAILURE in 5h 53m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"3cf93fb3ea3ea11b7cca05267c27dfcae3aa879a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-30 10:26:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/4/20240830134117/NEC-ISCSI : SUCCESS in 5h 20m 03s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/4/20240830134117/NEC-FC : SUCCESS in 5h 45m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d399c35a3bd670a5bafb32b386ea02b63e30e2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-10-28 16:27:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"849a22ae6fab3ea659ff8a556f758d694a41b476","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-28 16:27:53.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/815/","accounts_in_message":[],"_revision_number":5},{"id":"bd7f67361b1a8ff89ece43091b6c4cf032879d3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:32:01.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926295 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"848f692c9e16551cce3942d11ba1457bbc95ec89","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-28 16:32:02.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"e35de9cced6221c6b7f88b0eba0b5ef56a66fa46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:32:05.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926295 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"612e3ad2c027e997bb4f9a58869707ca1c542371","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-28 16:33:48.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"86c7fc1b5a2112d3efde3a584f422fa66be5f935","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-28 17:54:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5665401c1ee640199d082aea4ca1ae68 : FAILURE in 1h 12m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c8287ab249aa4cc2a9f823fa03d6739b : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":5},{"id":"18c9e57ea70075859ae2c9ebfe0db92bae001730","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-28 20:00:09.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/5/fc : SUCCESS in 0h 48m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/5/iscsi : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":5},{"id":"182538b75db002008ec31dc70cca370c54976442","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-28 22:26:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/5/20241029033638/NEC-V-ISCSI : SUCCESS in 3h 38m 24s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/5/20241029033638/NEC-V-FC : SUCCESS in 3h 49m 44s","accounts_in_message":[],"_revision_number":5},{"id":"97986d60301130df7b45d1da1bb8bf6e51132317","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-29 00:24:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/5/check/DellEMC_PowerFlex_3.6/7c786c1/DellEMC_PowerFlex_3.6/2309 : SUCCESS in 2h 27m 40s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/98/926298/5/check/DellEMC_PowerFlex_4.x/a7c8358/DellEMC_PowerFlex_4.x/1189 : SUCCESS in 1h 48m 06s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/725725e/DellEMC_PowerFlex_4.x_NVMeTCP/1102 : FAILURE in 5h 53m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b4bc487027ce2b6f0b479cde9fd799093c89b2e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-29 05:08:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/5/check/DellEMC_Unity_FC/797cf52/DellEMC_Unity_FC/2803 : ABORTED in 3h 00m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/5/check/DellEMC_Unity_iSCSI/8b4800b/DellEMC_Unity_iSCSI/2787 : ABORTED in 6h 00m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f33f4859a1e4bd680f59e482c0c09f0a94c12132","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-29 15:37:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/5/20241029184855/NEC-ISCSI : SUCCESS in 5h 21m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/5/20241029184855/NEC-FC : SUCCESS in 5h 48m 54s","accounts_in_message":[],"_revision_number":5},{"id":"e93132d4b8f4b909d2395354e181f9aea2d6d016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-12 12:49:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e67dba9a38d57627b1dec9507cef2cb41505bb28","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 12:49:28.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/906/","accounts_in_message":[],"_revision_number":6},{"id":"ede96a9900bc6c51f5b25be3b510958b7e3c1a01","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 12:50:24.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"ef7f26f78e0918c1a78120f88c8a0675bd99d563","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:10:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff162d4a4a74459f93d882802b61d088\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64186fa44d0d4f518df088bcf78b0ed5 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96cd3d70054646258044e0714b12e6ad : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fbdafc8975e99d25a5f2f8033109c502f062963a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-12 13:23:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2fb7f139b5f81f301a76486523c70605c8a95aed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 13:23:59.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/907/","accounts_in_message":[],"_revision_number":7},{"id":"83173aeca4a85a328b18d4e018606a5189b1da80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 13:24:54.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"32cb10041f573b83c37b45d208d12e5efc9f29c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 13:51:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279703069e084382b09000c7c6d0afef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f1cd55d5a24ee196c148124a4e2f5c : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed7f45b2bcf043648634d0dff641b938 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b0e3ac4bb67376762bdec74e407703c4867c20e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-12 14:32:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/81631569520a464cba65a75830965db8 : SUCCESS in 1h 07m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3035dae17bc14c64ac7df4302d539e8b : RETRY_LIMIT in 7m 19s","accounts_in_message":[],"_revision_number":7},{"id":"8d7d3efe324c951a929842b73f16872842b0a850","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-12 15:17:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3bbbd97e89794623b47cbab23e9a4133\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8379c027d534abe915dee68700d5047 : SUCCESS in 41m 14s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/255744b2965e4e7fb1ba565743a982df : FAILURE in 41m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0601ba161bfa4ffebf477be204db0346 : FAILURE in 1h 53m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5d041579fbe462f9047380ff44d30a7 : SUCCESS in 42m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb8fc55544db4ab0a2c38bc68cb2ea76 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6237dcf5afca487481b2e68a756a752a : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6b50d3c7c8e4463890f4b8eb7fa9008 : RETRY_LIMIT in 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4eee8f8635084017b1c6b3e267823875 : RETRY_LIMIT in 32s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/d... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"9cbc4ef3854f566968775562340ca716e2faa9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 15:51:01.000000000","message":"Patch Set 7: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4648e2a08d7248feb2bee871689f0123\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ccdaae164a4101a2cce8400579350b : FAILURE in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e890633e0e34a6091286871e8192288 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4e089e4e65f4ccc85d811a4a22a01ec : SUCCESS in 14m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1394b3de9d5146e6b70fcd5128241633 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/8578c297e62d4e0ba5ea92964859a14f : SUCCESS in 45m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cdc3b26b90574a77a7bffea8fc739764 : SUCCESS in 1h 00m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95ba81aa464346a2a339ee20b07a3cc1 : SUCCESS in 1h 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/313b14f23b3b4f0995b22b1f844da429 : SUCCESS in 54m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40a2af8e702149cd91c225f8acc315a5 : SUCCESS in 15m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/06adc199c3d84f7ba9328784b6f3b0fa : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb4a1f1ab5a144bab4834920b035e570 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cac187c9fb3401ab550ad149c7e99f3 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6c18740749f4fb8905da1e437f3da0f : SUCCESS in 7m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f31bfd573f2414fb2d2521ee844a647 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2d4917fb8e94b3698cf5943abe452bd : FAILURE in 20m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3627f357b51744528d45571fc72736e6 : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ea3cd4bbc274af1ad05645874791acf : SUCCESS in 1h 38m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5321899a43a34060b867dc6dc94bb278 : FAILURE in 2h 24m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b61999299ee247cfb50c1e4388898fa9 : SUCCESS in 1h 25m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8a3a9fa508e449539ddbdf26d6091566 : SUCCESS in 2h 13m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a170b6b1eaab4edbbe4fb9d2f25c76e8 : SUCCESS in 38m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d53b45acb05a4533b6db47d162292952 : SUCCESS in 57m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45620aa105eb4408a3c4fc9b230c4a27 : SUCCESS in 1h 34m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9819a1e42be54ca2a204783e0751bffb : SUCCESS in 35m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e7e458d08614988806e5973a39c1d37 : SUCCESS in 1h 05m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ed867922b66d4341b8b627870779b7fc : SUCCESS in 1h 21m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5948f78853d0474cabca939bead041bc : FAILURE in 1h 51m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0be14e7a7ac43ffa20eed8dee4cb90c : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":7},{"id":"fdcb12bc46006921f889a3a72b4b82df6dbf14f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-12 16:29:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"012837c0b138daf4e206d67d08cdaf8a5f377c8f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 16:29:49.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/909/","accounts_in_message":[],"_revision_number":8},{"id":"2e92307b1978fc2b4f39916a4ac89084c9ddf66f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-12 16:30:44.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"8dc793d9863cb9206cd3d934673bfcca514e7f9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 16:51:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f8bd9e813a462a9eab9a00d35e4444\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36ba89d86c254468ab60fb8b02ef39f8 : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c8381b1498b4e3090c4519f68a6fa28 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25821f9be6cd03e96a6e68732530876cb0eaf7a0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-12 17:45:58.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/db10132da7144d51a2e48b719f6583df : SUCCESS in 1h 14m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/737be9fe85b6465bb3fedaaf38eefb96 : RETRY_LIMIT in 7m 23s","accounts_in_message":[],"_revision_number":8},{"id":"b591aba920cb181f50e0190f70594422ed360f6a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-12 18:23:08.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7bbe5cd9995647f79bf64249022d7160\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de3e1ab0f519425f886b485d15b81aa2 : SUCCESS in 40m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17bab8d805854b86b595b555b2e51122 : SUCCESS in 36m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/066e50a41af941e6be6839bc1cf7107e : FAILURE in 1h 52m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd7444f5e57d44d08d53ac94a97ce323 : FAILURE in 41m 56s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee7c9d6c822d42ef9db49665e7ac6241 : RETRY_LIMIT in 32s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cdd3d9518091485897e9ec9779e4a9d4 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7bed261537446f9a7f8c799caeaef2f : RETRY_LIMIT in 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/045a081f7a854a988f8bb8f96c94aa9a : RETRY_LIMIT in 30s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-d... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"965c8e8705aefa3001e92932efa08395f7de8e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:39:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56609ced829e4edeb816923dd1e99d38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b9c0e7c1ac4cc9b0731e743a771eff : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d675b73e824976ac62c3c830c8ec08 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c1b31eb89214908a9f5aa5df28d6189 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e052a6a148143db9cb8f45ee0e4af7c : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/65f9d4e479db4f3bbebf818a048fcc38 : POST_FAILURE in 59m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efc77ebaeab741e7beec20a81506a14e : POST_FAILURE in 56m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/790200c32242485a83d432dd68357aa9 : SUCCESS in 1h 21m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e42664820934da7b5023c665ae2b202 : SUCCESS in 1h 02m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f589be5a42e94c5cb6b9f4a257d35d63 : SUCCESS in 20m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f8a1e9c7d8694ef993e0744299ab9801 : SUCCESS in 19m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/622d728a8c7d4932b9e7852c57bb69d0 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/528c2b73004b4fd2878c2afb5fc164a0 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c563d8004934fbb9439311f11f49a4d : SUCCESS in 7m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/df05d245f13d4b34bc14afc64d2a3f65 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac3edaef31b04c4cb5f136d9b10e901e : FAILURE in 19m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10c8ee7f781846f4aeaf6c5b0dfede3a : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0883da0b82b945e6a8c4da53e0711c23 : SUCCESS in 1h 35m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/952dd751677f4439bdacecec615b239e : SUCCESS in 1h 40m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/212f1281df3e4cacb6c10f9cb7f42f66 : SUCCESS in 1h 27m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9560e9719fa94ed2ba2e16fcc85fdd20 : SUCCESS in 2h 04m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/54f0ad00901c4104b936d55ff8b743f6 : SUCCESS in 21m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f398e1efdd874386ae4acc4e79d0b7e4 : POST_FAILURE in 59m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/06bef522eba84e18bd5ea8dfbf4deb1b : SUCCESS in 1h 50m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d1412b69bb32474dbf47c733d6b90e82 : SUCCESS in 43m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd8c097f440e4a6fba130fb2ce0106dd : SUCCESS in 1h 37m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a7cd7378f86b4c22be7dbf0fe1661a36 : SUCCESS in 1h 29m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2d70ccc23b24c1584a43759c05bd94e : SUCCESS in 1h 40m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07720e98f225405b828e1847418cc5b0 : POST_FAILURE in 55m 18s","accounts_in_message":[],"_revision_number":8},{"id":"5d2b45b52526e4c4ec1d61de3c4053be1a4034a3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-13 05:07:14.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/8/check/DellEMC_PowerFlex_3.6/42ae017/DellEMC_PowerFlex_3.6/2420 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/98/926298/8/check/DellEMC_PowerFlex_4.x/4f55a1b/DellEMC_PowerFlex_4.x/1268 : FAILURE in 2h 12m 42s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/8/check/DellEMC_PowerFlex_4.x_NVMeTCP/03f7b9d/DellEMC_PowerFlex_4.x_NVMeTCP/1162 : FAILURE in 5h 52m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"74b6eae3ef6d9be54623e7ae5ff7cd4a8b44edc2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-13 05:53:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/8/20241113104449/NEC-V-ISCSI : SUCCESS in 3h 44m 15s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/8/20241113104449/NEC-V-FC : SUCCESS in 4h 08m 34s","accounts_in_message":[],"_revision_number":8},{"id":"80025525140c463fad1082b523d2b041d1b547f3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-13 07:12:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/8/check/DellEMC_Unity_FC/55b98df/DellEMC_Unity_FC/2873 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/8/check/DellEMC_Unity_iSCSI/f997c7f/DellEMC_Unity_iSCSI/2841 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"09cbf06338c34dbb47c445e3a91b394efedb660c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-13 19:02:45.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/8/20241113191512/NEC-ISCSI : SUCCESS in 5h 19m 30s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/8/20241113191512/NEC-FC : SUCCESS in 8h 47m 30s","accounts_in_message":[],"_revision_number":8},{"id":"6713d018bed6cf16f9c96c931a26d5c2fe45e01f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 13:19:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/850/.","accounts_in_message":[],"_revision_number":7},{"id":"3793bd165e2d9eecabb3ecd64bac97f2bb05d6f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 14:55:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1416/.","accounts_in_message":[],"_revision_number":7},{"id":"c49d38248ba6cb2fbaae76fb38fdfa33d08be53b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 15:10:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/852/.","accounts_in_message":[],"_revision_number":8},{"id":"6826961782ba375cb76e00d92999101376af7664","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 17:57:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1418/.","accounts_in_message":[],"_revision_number":8},{"id":"31c49f2ce55bb03b00c89494f49797fa835749ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-19 17:12:09.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"168bbed234951d2b4779b8a34aa8267b36c85570","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 17:12:18.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/947/","accounts_in_message":[],"_revision_number":9},{"id":"024be7b449c434c0859014f8157c6edc4998cee8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 17:13:11.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"1296f10d968582695844c463b0d6fcead50f6eb8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-19 18:14:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":9},{"id":"a8c5131e47e70bcbf5f2fd55090141f260ca76bc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 18:14:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/888/.","accounts_in_message":[],"_revision_number":9},{"id":"f81c7b4769f20c46990f064a07f5e0f5fd0f1868","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-19 18:39:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/800dc9b1a2034cd3b804728e33f612c2 : FAILURE in 1h 20m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1e82997fadf9461497dac25f56a2ed0a : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":9},{"id":"6037e08d1926cde60132df2d75269e2b44aeb7a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 18:50:45.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1453/.","accounts_in_message":[],"_revision_number":9},{"id":"7bc22944e0f56b25acd44736315c621d2e8f878d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-19 19:26:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5b52d33c46064852a34dbda6c4ed4122\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bacc146fbcb744328eac02042c0eb766 : SUCCESS in 48m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5201bfa8b6dc4dffa3e6950eb738f64e : FAILURE in 2h 10m 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66e882a0bc6342d6ac899e92a8b4bc20 : FAILURE in 1h 31m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a33bcb3dae9042659c84907ab39f660e : SUCCESS in 39m 01s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37383d16b75b41a48859b1e1d1d227ab : RETRY_LIMIT in 34s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84a2c2f67fcf48a7a5dd40e0c7d2f0ac : RETRY_LIMIT in 38s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25113a4d3c054631bf898afb655eca1e : RETRY_LIMIT in 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d47256f5732142daabd6dcc8d1d09934 : RETRY_LIMIT in 32s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob ... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"d044aea53c94e74707cd36e55ddef33217141840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 20:12:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/929b5d96b4a84ae9ac8a89f5fb3a7187\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbac4b16d8bf4772a3edab2ed922c02b : FAILURE in 22m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8032396aaaf443a485c03ea83b0f6089 : FAILURE in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"64f37a4a7c1311c28e32bc962f4275b704e55a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 20:15:24.000000000","message":"Patch Set 9: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0092c0214114a079851ce22b04e4a11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f930d976abc84723a7ff5deebaa806af : FAILURE in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d15d8ab0ed9a4cc9affb372dc14d28f8 : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fa84789dba748d5a7872829b4a777cb : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7d28a45e514f67abc51fe5d1e275e6 : SUCCESS in 14m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/96ac229f90454f84885206b549e0ee13 : SUCCESS in 48m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21af6ca4d0d146f88fac3b32ac3b9cfc : SUCCESS in 56m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cfd237c5fd94c5e9b6a31499491f2b0 : SUCCESS in 1h 14m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55ca5d064b75434784481005f6f0e800 : SUCCESS in 53m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5fb4e8ce1e40be8bc1d22a851dffeb : SUCCESS in 19m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9b2ccaa14064688a09041171dd119eb : SUCCESS in 16m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4569c08140ed4e1d9f07679d444f06ca : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/980d1f615fd240899dea560c62d833a5 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e214a4b7e1744f87b42979aa52854bad : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28d034d738de476ebb4469fc798d5c95 : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4cad6cf7921e4573be1884c079953149 : FAILURE in 35m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9239629c09e416bbe7e6d46fd446e77 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fe621facaf940f5a1770d7dde7453e5 : SUCCESS in 1h 51m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c55cd25d44944d06bb0643feb6db25b2 : SUCCESS in 2h 23m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/883672fb8b8741c4a88f857910d6054a : SUCCESS in 1h 52m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e3dfb23e25a04c6a9ecba1a6bdc92e19 : SUCCESS in 1h 56m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/de215f4f7d5a4ced971e0c4bf6f14272 : SUCCESS in 41m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/639c25ddde5d41d49bc376063d216b3f : SUCCESS in 1h 00m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d2fdbfa0564d4a83bb93cb0cd318e787 : SUCCESS in 1h 26m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2568ae546344232ad52aebbcf177b9f : SUCCESS in 40m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df8d08159bfa4fc4818deb2aff9daec4 : SUCCESS in 1h 18m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/75f2350d263643e9987ac13e9f1471d5 : SUCCESS in 1h 56m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a140491f8c7e443a8faf9edf4eb95a47 : SUCCESS in 1h 28m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0514e249e5346a3975ae426332d8213 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":9},{"id":"06bd8d416088c814df1dffcb4e5e577940b96eec","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-19 21:00:13.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/9/fc : SUCCESS in 1h 01m 59s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/9/iscsi : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":9},{"id":"4ae24ea9da25fb767558b763c9d29ed4a2fe38ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-20 14:07:37.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/98/926298/9/check/DellEMC_PowerFlex_3.6/ab53474/DellEMC_PowerFlex_3.6/2455 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/98/926298/9/check/DellEMC_PowerFlex_4.x/f7a4eef/DellEMC_PowerFlex_4.x/1302 : FAILURE in 1h 56m 41s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/98/926298/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/c6f4f75/DellEMC_PowerFlex_4.x_NVMeTCP/1194 : FAILURE in 4h 37m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"bbd92d82d63aa682f43f6b633adf798095413ce4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-20 18:44:52.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/98/926298/9/check/DellEMC_Unity_FC/4aef747/DellEMC_Unity_FC/2906 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/98/926298/9/check/DellEMC_Unity_iSCSI/ce8bf29/DellEMC_Unity_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"b3c0cf21b016777005e41c459e847fa666a1794e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-21 05:15:35.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/9/20241121054333/NEC-V-ISCSI : SUCCESS in 3h 50m 30s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/9/20241121054333/NEC-V-FC : FAILURE in 8h 31m 59s","accounts_in_message":[],"_revision_number":9},{"id":"d1f01ef9b34781fa1c248325e397c890f80b4d79","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-21 08:03:10.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/9/20241121111640/NEC-ISCSI : SUCCESS in 5h 18m 24s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/9/20241121111640/NEC-FC : SUCCESS in 5h 46m 27s","accounts_in_message":[],"_revision_number":9},{"id":"72d891625b74d9c9a31467de0d84115db6ddf304","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-12-16 19:36:24.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8badf9128b65d13f7514b60f9c7698617dae78e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-17 13:58:56.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"265684bb08d242839e3fb767ca335ed9e74d40d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-17 13:59:07.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1335/","accounts_in_message":[],"_revision_number":10},{"id":"7b4d9c22783bf8c7d9264072ea96a35e12333182","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-17 13:59:09.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"cd5a9dfdef7ce36da006ad057fa54801cb271b37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 14:00:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4849333cb5864819b6b70f840b904c9a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c1adb98e6c64abc9a4ae448de9a37e0 : NODE_FAILURE Node(set) request 300-0026357916 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c88cc1b557ff4bc19615186d266323a2 : NODE_FAILURE Node(set) request 300-0026357917 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d6753fe05a829a6323a3e2b3da849032804a23af","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-17 14:02:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9aa5c19b23ee6fca3171223836b7d47950e77844","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-17 14:43:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926298 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":10},{"id":"f43f0388c9dd6d18424be312dac7f531edf8aa06","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-17 14:44:38.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":10},{"id":"f7daa1a7d4559a51636ea0914dec4718f15cf0d7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-17 15:08:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3a4a38dae9c4b1aa8a5b64a4d36d62e : SUCCESS in 1h 08m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f910751f075f4d0c88bf7fbe13c26bca : RETRY_LIMIT in 8m 01s","accounts_in_message":[],"_revision_number":10},{"id":"d7224bb89e234802712824f322e1cb8b9bb4883c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-17 15:21:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 19m 46s","accounts_in_message":[],"_revision_number":10},{"id":"7cf1273c3e709cffd8da68f1dd2cc0be67b0bbc4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-17 15:26:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/542e1cfd76c847d7b2e2b0028fb9d203\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77096ecc89054ec080582059b5833bfd : SUCCESS in 47m 52s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2ab3cfeb9764ad5adb5690deadafa3a : SUCCESS in 38m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/636577a73a884b0eb1ac04c029392a5b : FAILURE in 1h 26m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62d2bca69a2543bfae06b1f77db40639 : SUCCESS in 39m 07s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d297a855a7a74bba93099f5550e344b1 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08b44c7c22464409956809d0be89b928 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3662f3409464e7c843afdfe6f72d9f0 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb362f426a3e42818f900bb887718d28 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#511\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#468\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#511\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#468\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"e3053fd706789d2c9586c1e8db293f8d8e3003ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-17 15:49:50.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/360/.","accounts_in_message":[],"_revision_number":10},{"id":"7bd8c6e71992a63dc64b15ce01ebfc6fb1c62605","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-17 16:08:15.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-98-926298-10/FC/ : SUCCESS in 2h 8m 59s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-98-926298-10/iSCSI/ : SUCCESS in 1h 32m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"df2c5733c70482b3cac1e0d473f339a6e8e8637d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 17:46:36.000000000","message":"Patch Set 10: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb5e11ae9b7149fa85ca31def0ff1631\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938e7c6fa75c4af5abdd20de55e51dbc : FAILURE in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a60d237f5e41c296feb37d5fb683e3 : SUCCESS in 11m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bd8ba80ccd5492d9809585dadb6f4c5 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c146dfb9fa457d9999f2afe6d24c3a : SUCCESS in 12m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/78a383875eb648bb9e2960fcba6422b7 : SUCCESS in 3m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/60522b8493bd42a5b7e6e357ae67154c : SUCCESS in 57m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/767064359e6a428b99256d0ca6c43997 : SUCCESS in 1h 46m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/965b0ce4ba1742d79a88130d616f5971 : SUCCESS in 1h 30m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16abb0d80d334785b89428351ac4119e : SUCCESS in 1h 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bdfd508f6f8443e9a7184ad4ad1e832 : SUCCESS in 22m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7691a10c0e045a2a00bbb82c7f9c6ee : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f2ca809a62d44a5db4b1e5d53a6638f2 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02bb5a7960194f4eb89479a4565b3b37 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6255bafe7774402a81c8aadeec2c29a : SUCCESS in 6m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b0a7f0a71d5b4d6fa30eaebde70fd2a1 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cff54a035d3e4c068b6bf83dfc6d054e : FAILURE in 37m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d05a836233542b8b4b0f1908187a90e : SUCCESS in 20m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ebd3c0a79c84709b00463c8e5265494 : SUCCESS in 2h 34m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/30844a2b0fa447489b5481fe2927303b : TIMED_OUT in 3h 05m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bcd82a623df34a85b26c98122d8700a7 : SUCCESS in 2h 01m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f056c33a6d5b4c76af1af73df0cb4209 : SUCCESS in 2h 26m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a150f7edcb9e421e99c20df5cb4073cd : SUCCESS in 55m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d54f5695efbe41669687cffabcde2c4a : SUCCESS in 1h 15m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8fe7b8c094f45f0af65301dd7e6f534 : TIMED_OUT in 3h 08m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/251139bafbd643da9382484b8fecdb75 : SUCCESS in 37m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32e04128f0ea4f7aa6a108255dc92d78 : SUCCESS in 1h 20m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/92b23af02f7244fe8c4769f2b331040c : SUCCESS in 1h 51m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74e5c9623cbe40b7b02d566af5155129 : SUCCESS in 2h 25m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e9448cb4987472ead62d74cce894084 : SUCCESS in 1h 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d19be2509bb49fbb9824246332e2233 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":10},{"id":"59a1c2961cd64c4e1f3edd152afb21f6d7f98bfd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-17 17:57:13.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/10/20250217225858/NEC-V-FC : SUCCESS in 3h 56m 06s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/10/20250217225858/NEC-V-ISCSI : SUCCESS in 3h 58m 11s","accounts_in_message":[],"_revision_number":10},{"id":"03667be0b77dfc72332ee2d5d3b55373bc843307","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-18 05:59:13.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/10/20250218090527/NEC-ISCSI : SUCCESS in 5h 25m 38s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/10/20250218090527/NEC-FC : SUCCESS in 5h 53m 43s","accounts_in_message":[],"_revision_number":10},{"id":"32aad91a7e44ef91435f2b50aaa4a9ca12d2600c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-18 08:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/10/fc : SUCCESS in 0h 33m 06s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/10/iscsi : SUCCESS in 0h 48m 32s","accounts_in_message":[],"_revision_number":10},{"id":"748effefd211fa4f48e346776ce2d4947f8e519e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-21 19:19:32.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dad97e0a6da3a2bd1abec2e43735b98ceb3a99c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-24 09:54:33.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ffc6bbc4189f48292027f57837a1f55e9836654c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-24 09:54:41.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1354/","accounts_in_message":[],"_revision_number":11},{"id":"90c8392422e3ac1858317973686689d4ea99080c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-24 09:54:44.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"32857d46eee426f66d73773de6b7591fa924e995","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-24 09:55:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7f9faca5d9f08686678b281ed66440fdddc5989c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-24 10:09:44.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e368fe29f4e54f3cac12dd927d7fe8d8 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/146b676bd7a246c2bbb0c8c3940f5a47 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"5b0ce6e957c1c715f815712f403ba1a602aa190e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:21:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ab22e28fc14104a4f651502eb63917\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b09b03ae4be145c396b1355872ba3a38 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1dbb364157514d77b816da4b129e2d7a : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d005f8a447f03283b520ea493c79037c1332a8c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-24 10:42:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":11},{"id":"0bc7bebccbf8aba68641c6ff34648ba67dcb9c1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-24 11:29:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 07m 15s","accounts_in_message":[],"_revision_number":11},{"id":"5a78b3d40ad4b17ea47c92590d6453689d634af8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-24 11:33:42.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5636833f9da24635b2229352f6cca9fd\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/698fdb0659da4e73ac412c0231a56de9 : SUCCESS in 41m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/12e16a4f19da40ce8160ad3ce9524851 : SUCCESS in 40m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cf1d111b8e943008a5d7e29918b0f08 : FAILURE in 1h 38m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbac31891b0c49fd9d8871cc9c7de8b8 : SUCCESS in 44m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71b642af443045e8bac082137ddffef9 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9533e9cab6854b0ea64bb488028d72df : NODE_FAILURE Node request 200-0000026480 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3e6f3425ba3046e0a6ace22e7cff5e53 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28548daecfcc4135ba387d347eb0b59d : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"b047464d38c1064ebaa8997cc1abef22e9ab230c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-24 11:49:03.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/381/.","accounts_in_message":[],"_revision_number":11},{"id":"18a4dc1d113f7a54344ae91fa42c489d5deb4ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 12:31:40.000000000","message":"Patch Set 11: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03760049fd1420ba64c3adb0a7f18b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9ffb20b23846938475b4d5830d21e7 : FAILURE in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bb621e4e34843caad19bae1beaa5541 : SUCCESS in 10m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7c91cac9a74420b95584234e103715e : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a660122728f44b23b635c3c7cd300cf5 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/0f67511055de4488b093ccc3f8e2f7c6 : SUCCESS in 1h 03m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dee94d5e576e436eaa01383eb68ca052 : SUCCESS in 49m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e32f093273f4bcf82b050423165d8ca : SUCCESS in 1h 41m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c25853d7c444da7932bc18a0b31cd4f : SUCCESS in 1h 03m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ee95d0a17364d81b9a538eeb81c9e3f : SUCCESS in 14m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8b86dce8edb94f37b297bdc78848ef7c : SUCCESS in 22m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1faa2de0700247478e664bea7b049e4c : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f19df7e61ff43778dd0e2f93f20068e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21ac19a99dc64406b2491b515fd9403b : SUCCESS in 5m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4131bc0f63344aeb905ad366032e1c0e : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/460949bd623c4e0091440f50880b730c : FAILURE in 35m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e96f48f04d4b45079cc759d49589f796 : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7df8c88aeb5b4524b0dc88d46511da2a : SUCCESS in 1h 54m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f281b366f89a44d1ae374f79baf0ff0e : FAILURE in 2h 33m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/730c745e8f944a64b1885c93f8ac5f63 : SUCCESS in 2h 05m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e7f922e2b069437081a3235ce358d3a6 : SUCCESS in 2h 11m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2495dc3e102c49c3b07a683b4d87b3b6 : SUCCESS in 25m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/274c349d95434eaa8fa82c703bed86ff : SUCCESS in 1h 07m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/054179c290c848439c83333942234685 : SUCCESS in 2h 01m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cbd13bee3d83449ea713154e6ee99a45 : SUCCESS in 36m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd8e4c4e11a544f08b6b98ae8047b53d : SUCCESS in 1h 33m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/45679b8058944605b8cd26a2470d1e3a : FAILURE in 2h 23m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8fccbef8c734d07a98e1634f8a1952d : SUCCESS in 1h 53m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7e43c092fc146ef94090155a7f5fa88 : SUCCESS in 1h 29m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb22267637634c269ba35df01f05451d : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":11},{"id":"b41a062719006a83bba3fff77fd079f83dcdf958","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-24 20:59:14.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/11/20250224221641/NEC-V-FC : SUCCESS in 4h 13m 02s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/11/20250224221641/NEC-V-ISCSI : FAILURE in 7h 42m 30s","accounts_in_message":[],"_revision_number":11},{"id":"4b5bb45155bf2d04680ce68056f9d23cc02fd1fb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-25 17:08:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d3f05a3b315c06a8a8210d64dd730c3e3ac363c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-25 17:50:44.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"2fc897368a508ef7bcdaab4c98709c45a902ae14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-25 17:50:55.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1367/","accounts_in_message":[],"_revision_number":12},{"id":"3283e4541a79e60bd1e162590a432d5ba112aa88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-25 17:50:57.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926298/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"99f410b68acb3aaffe1ccc1fd681d6b1d03ce8a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-25 18:37:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":12},{"id":"532f333b905cf9bbd77d2fa0c3a6d7299d0ca2fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 18:42:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09f0c66d06945a9944a04e3aebeaee6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60537b2733384f5ba5c03ffcd03832a5 : FAILURE in 28m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/151135184ac84474bce3e7e26d1be5a3 : FAILURE in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"33b21082fbe4203729f4e367cb0cb54bcd0fc7ef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-25 19:18:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 08m 16s","accounts_in_message":[],"_revision_number":12},{"id":"69c9410a7db3ab7a70e996bfdbcddbbf4a64f707","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-25 19:37:16.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9bc482cb1a594ceab725fc70cbdb30a5\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/58e46c7d17a24867915ed7673f5031c6 : FAILURE in 56m 07s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8069f331d2f34623b01c92d0822fc5ba : SUCCESS in 47m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bdddf958e9154dc0b0a36e6cda44536f : FAILURE in 1h 45m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98c2a8c908114a4e9e75c0b00468c77e : SUCCESS in 46m 10s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98f4c6208062442ea8844f8bb7c1538b : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/938bd430d2b74282aec94557d06cfe5f : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fa9524653e94847881efb608193ec1f : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b79d808b14ec452985f67a4e01b27129 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"4010f880310deea02b3570e0be17b1d9e6a135ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 20:08:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97368154b82d4b39989940982e552b70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6498b495480d4abdbebd08e21474ba04 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb28c14372b84539ac6f02939322a3c8 : SUCCESS in 11m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87f83aca7f6f4bdeb1d5c17f0effb0d3 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48416c8c9c849ea9210ef6c3528a8fd : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0c310ef112f34c79bc13855be987cee6 : SUCCESS in 56m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd65ac347cb54a31903530f95dfef4f8 : FAILURE in 1h 14m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16ad1475944f4d4ab85bd70326086487 : SUCCESS in 1h 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0400b6602454538a064ee22b0a1436b : SUCCESS in 1h 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d465e5390e54cb0b476f2ee6249d39a : SUCCESS in 14m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdc9734f5eb347f2b8b16bbc5b534b41 : SUCCESS in 24m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/44a7deed418e427aa627b5c6bcb629c9 : SUCCESS in 7m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5271da2b78d48339eb6df7338f3455a : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2641341737b49e9902bcc05cb82194e : SUCCESS in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3e27d86964a44ae88f08f6085de521c : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f201242a858a4289a7665f4f7e6d077e : FAILURE in 29m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91e7a9e286da4e73983f0887f588b870 : SUCCESS in 7m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f63af10801343f8a24ae5a4e44e709e : SUCCESS in 1h 36m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57809df35c3a459abb9379f36ac2306c : SUCCESS in 1h 47m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6860470ed71464c8bedf54b548fed7c : SUCCESS in 2h 01m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4385c0339aaa4d17a9ddc8debb7fb62a : SUCCESS in 2h 05m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/56b6912d50444ba0ba3a1690ff5bf697 : SUCCESS in 43m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ac0fc39dadb742ccbf029deeca371178 : SUCCESS in 59m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d4f297d949244548032eb6b39e4d323 : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1632b5b233d64ad6827483089e2c1ddc : SUCCESS in 53m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/301a78abef2445b08e7d8c4f0c71ebd2 : POST_FAILURE in 1h 57m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d7a1f309ce2d4600a89a3b8ae00d3cbb : SUCCESS in 1h 42m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/712b077d1da842e48ce6c7bcd6e97543 : SUCCESS in 1h 25m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2db01296ad90435186a3c3e24fa5a058 : SUCCESS in 1h 29m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2a12dda81c49a9a88f7335e80aa8aa : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":12},{"id":"c7c8ce4b435c311a32668a452cc9d0a846693641","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-25 20:19:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fd8ff7f6230f37fd0c424e6a3e814241c1788c32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-25 20:26:55.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2aab213e3f884867b941f76c8788d69f : SUCCESS in 1h 17m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6765438ddd3349b0863dfa1a6df9045e : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":12},{"id":"ecc36c270f7f6e0020597a8714309ae535b95b52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 20:44:13.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcc7667f5af45db860f2ff2e8196655\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f704c999ebec4632b5a0192f4eabf7d9 : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/221966c7c4ba47188b324df533e1d8c0 : SUCCESS in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"070d6f6f7e0c708605e79f6221bcb08c62df597b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-25 21:40:42.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/396/.","accounts_in_message":[],"_revision_number":12},{"id":"a073a08d422179d0816e851c606990ca30e6dea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 22:38:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66265343c13842fb94cc00159758038b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce27434e0c840499410c93b5c620371 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef9cea813134e61979dfad78392c508 : SUCCESS in 11m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/596cb7b61f40439b878ef0e503bf9136 : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b73c691b4f248b4a0976d33d578ba60 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4be90241bb9b448eac2efda6c1eef93b : SUCCESS in 55m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/55de589df1434ad689f712bfaa037a49 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b87bb0fc46d46e1bf49f6666233d4de : SUCCESS in 1h 37m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a3a4c40ccdc4cbb88312f2b72746610 : SUCCESS in 1h 03m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/983904f0d8894fcdb1fa323a34035db2 : SUCCESS in 14m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03715ce062a84ed997e41537f719379c : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/10fe1a8e01cc432ca425921bcade3cc4 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c71d2f909f524661b4c668e8d3570750 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9f2506c8fe3849d195e35e1d09bc9399 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/aa4013b1371e4012ab56b647363da503 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60a587764e9b4cea8c7e1d2434a4f882 : FAILURE in 47m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a358a9a5c4d44e5abba3b5a6676ce31d : SUCCESS in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83e49b35c8614c25883186a0b16cf181 : SUCCESS in 2h 01m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60a8a1f411c2439493041409defbb352 : FAILURE in 2h 15m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35574daaba8c4263a9bee2539264497a : SUCCESS in 2h 02m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/089ae2907fe94332b2c576f93ad726cb : SUCCESS in 1h 31m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e71132fb94fb45a2b77217ba004f7fc2 : SUCCESS in 34m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84b2a4acc35c42fe98adb4db3bccac8d : SUCCESS in 1h 12m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/11ed5ce89e58496896a01a71ad26069f : SUCCESS in 1h 19m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7b7250d484a745138b03d801e83e1391 : SUCCESS in 31m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f637b35d72704081941f45ea63ccc8f8 : SUCCESS in 1h 15m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d6ad00363c19442caaeb1b7bf4b2b626 : SUCCESS in 1h 21m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f23921da3414c1db84ce53edbd13d66 : SUCCESS in 1h 21m 07s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f73261eb1a4baa81e35bab98e8fee5 : SUCCESS in 1h 15m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d1b4c57e87b473db989a72e738c98a0 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":12},{"id":"95e12c1736eb0b5624a498088642292333ae5aeb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-25 22:39:04.000000000","message":"Patch Set 12:\n\nBuild failed.","accounts_in_message":[],"_revision_number":12},{"id":"c8488dedefd784955577fd38589428448bc2ed70","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-25 22:39:22.000000000","message":"Patch Set 12:\n\nBuild failed.","accounts_in_message":[],"_revision_number":12},{"id":"b202a482223886857b8bd3a3a99b66f890e4b24a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-25 22:39:56.000000000","message":"Patch Set 12:\n\nBuild failed.","accounts_in_message":[],"_revision_number":12},{"id":"3117282d39f5f3d89c5f688124d1377a3bf1d262","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-25 22:40:16.000000000","message":"Patch Set 12:\n\nBuild failed.","accounts_in_message":[],"_revision_number":12},{"id":"8c357c7caf15c7a2127df2d9d9d762583c7b3e98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-25 22:43:31.000000000","message":"Patch Set 12:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_02_2025_22_38_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"843b0283dfb5fff2e4a36598bcea635c064e7826","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-25 23:39:06.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-98-926298-12/FC/ : FAILURE in 5h 48m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-98-926298-12/iSCSI/ : UNSTABLE in 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"2687c98ecef8650d07ed7bfd116ae71bac8fe672","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-26 14:59:09.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"df35cd5df7de49d988cdcac289787f60a559215a","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-26 14:59:19.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"578de496018991c50d5f6fd619965df6e7a16d1e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-27 04:54:04.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/12/20250227061956/NEC-V-FC : SUCCESS in 4h 18m 59s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/12/20250227061956/NEC-V-ISCSI : FAILURE in 7h 34m 04s","accounts_in_message":[],"_revision_number":12},{"id":"dcc0e34d6ae636d9c087e1b91f2fd4eb35c65779","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-27 09:00:08.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/12/fc : SUCCESS in 0h 58m 09s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/98/926298/12/iscsi : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":12},{"id":"b44eccd772d1ee649735205c0af4e6c7d34d81b6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-28 08:38:28.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/12/20250228084617/NEC-ISCSI : SUCCESS in 5h 30m 42s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/12/20250228084617/NEC-FC : SUCCESS in 8h 52m 08s","accounts_in_message":[],"_revision_number":12},{"id":"31038ab2524a0af52d28aeae1aa895414a869e96","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-03-04 14:14:10.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f83f560edcced334fb11e05615e5f81c63511ea1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-12 14:29:20.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a635a188ea0c9b77109374b7cd18e6e3a3837891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-21 14:36:55.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"31d3e395748078674fe8bc1d3951580599a0c4e4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-21 15:05:47.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/889/.","accounts_in_message":[],"_revision_number":13},{"id":"8997f3d3c397cfd1b6ac4739720a4c9138711c5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:29:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47542b2b80f4400fb8038490f6ff3e46\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02425fce9a5a4214a8746bf8f25892ec : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2647ec890a144a678baef19ecd89398d : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c30e965d0a844d1beabb54226357907 : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f2b3784679f2e657ed51b595a547ffa16dfa23bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-21 16:12:52.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/767/.","accounts_in_message":[],"_revision_number":13},{"id":"54de085b5f2a5f2bee82b5267131cf618c71eee8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-21 16:30:38.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":13},{"id":"9989cac481370d614e046d4eb38a7423824efec6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-21 16:35:22.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/47e717a8f6b140feafe2ead03ab0d83d\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/78094a932a9b4ce8b8fa95aa86656e87 : NODE_FAILURE Node request 200-0000053952 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f8b5800df714e17a58885d70077f2e2 : NODE_FAILURE Node request 200-0000053953 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bdcf44ac73cd4ae6b4673d3a49229e5b : SUCCESS in 46m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a0c4723a33d4d83a28388c48621c5de : SUCCESS in 46m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d1705d828d4945938747bb38329af974 : FAILURE in 1h 24m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a8cdc5291e046388744e96f64f477cf : SUCCESS in 50m 05s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d05bd674b1c946a78e1734815715ae73 : FAILURE in 56m 16s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/282a734fed554473a66ceb1d29d85cd2 : SUCCESS in 56m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5b051a3c74c4a478972aa530e80aafb : FAILURE in 1h 57m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d5968c4839c49c296e33d2c70631861 : FAILURE in 48m 17s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None ... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"50c7550bcaf83ef9e3679efc58b2817ea561c1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 17:46:00.000000000","message":"Patch Set 13: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a309087bcbd04fce9bf2957f4ba78bd3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d19c3e2756849988d362af5eab6ed70 : FAILURE in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8675ec1726304abc8740af813677f9ad : SUCCESS in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77635f1652254458923218aff1f54459 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/505a508099f9495698d7def0e1be0c30 : SUCCESS in 17m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf1d6e9925c4e21b04688a97f7954e5 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/a1c6458308b447258cc86ddc5adb2150 : SUCCESS in 58m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d16d8ac0f96a475c844a128cccb1a1a0 : TIMED_OUT in 3h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b694b670cb14ff0a4e4a5b25e2c613b : SUCCESS in 1h 30m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8bd82f227984a13a0674f4c24708cae : SUCCESS in 55m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673119732a644e538cacd31376f552d9 : SUCCESS in 22m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/12f34e1d256b4d709f585d4395f572b4 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/466fea2cf9284e5d96a5bc055263fbda : SUCCESS in 6m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d1a9abd750b49a4a6bb6fbe37d3ea57 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79533144404f4389a60a19449701b7a9 : SUCCESS in 7m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b62cb6b14a344df82650981df8340f4 : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5fe1264cd7e34c318dc76e569f5e22f2 : FAILURE in 40m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27ee0cd536734436b61c3585c7ce0518 : SUCCESS in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1eb5de1872346b08b1e2eb744b6df0f : SUCCESS in 2h 16m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23e115c83e1949f4864aa26cb1287808 : FAILURE in 2h 23m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8985fd26e1b24ae2897e3e189cb27993 : SUCCESS in 2h 00m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e839b7edb9f744639effe6c67e194b6d : FAILURE in 19m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e6ff42c62ad44848a422b2343039e6b : SUCCESS in 37m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/18b5ee11e27f401d809b0f57d8c5a2ae : SUCCESS in 1h 13m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca45012f2cec4186a4f5b33bff098706 : SUCCESS in 1h 36m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/23e61f5f4f944bc98a16c92904f10a7f : SUCCESS in 53m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/910bd29274524596a18ccd0ac76c38dd : SUCCESS in 1h 32m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/37a4cc581bc64b60908940e16e5be278 : FAILURE in 21m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c99bac2a16c46c099277dcbea6e8c22 : SUCCESS in 1h 33m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e86692c3d9c4417b5fcbd481f4cfbc8 : FAILURE in 1h 22m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1d5f74b2a524f229f5bef1c3566c9c4 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":13},{"id":"a4b9cf332f4752a7be3e64f229b079392cd30620","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-21 18:02:06.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 18m 50s","accounts_in_message":[],"_revision_number":13},{"id":"5bfef8d10097fe5fcd4c82f7a2bbcef00cbae391","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-21 18:17:17.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fe77c2a4160044dab676d1dfeaf668c3 : SUCCESS in 1h 19m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/13e4f155c531483695807d3eebf7b527 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":13},{"id":"444d0ef361f80a131b263f837faad65df8527e25","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-21 18:40:03.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"b431750c67a42294aff2063a834d9dce2a3e3046","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-22 15:54:14.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/13/20250522200508/NEC-V-ISCSI : SUCCESS in 4h 35m 28s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/13/20250522200508/NEC-V-FC : SUCCESS in 4h 49m 03s","accounts_in_message":[],"_revision_number":13},{"id":"56d35811ff7a60cec9113492ba727754eba2b7a3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-23 07:53:25.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/13/20250523074425/NEC-FC : SUCCESS in 6h 10m 07s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/13/20250523074425/NEC-ISCSI : SUCCESS in 9h 08m 57s","accounts_in_message":[],"_revision_number":13},{"id":"e0637e90691e5038f1b8d48364fa3e6c20a06590","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-05-23 15:29:27.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"f56b0e8f156885208aa21d09720e55d399acb5ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-26 14:33:30.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"a57a7f7bb1a850fc23087dba84bffdfb636b6a21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 14:50:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d974a3debd41f1ab878cf3b2d06344\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9def04b0d5f6424981b6c0fcf89551b9 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/44f95c81df354e79aac869000c5b9099 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/792d9f00f1b34a959b6dba7988185cfc : SUCCESS in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f61d147e63cb879dc5bd0789e8e2036166e6f647","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-26 15:29:31.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c5ba8ce3ebfa43b2ba4e48dc5d65f071\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b8c3de71d19436f95c3ed1a170802cb : NODE_FAILURE Node request 200-0000054554 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9dd16f2bc2e14398a9bde5057ba8f153 : NODE_FAILURE Node request 200-0000054555 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90c572b7d28b431581bd243235c5d604 : FAILURE in 23m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee2f1b47e7bd41b09f71df51a8ac5ac8 : FAILURE in 21m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02353755d7a14189813a29e17698dc86 : FAILURE in 25m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/329c10d649094dbb929d5861b6dafca3 : FAILURE in 25m 54s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cbe70e126e14290a4b20739703c8225 : FAILURE in 28m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7510c9a719da486db0c4dece023fcf82 : FAILURE in 30m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1953df2960ef4959835841ab34ee9427 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d15860793fe48ef841c2b2163697cab : RETRY_LIMIT in 33s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"e1a62ad84ddcc5e7c3eb292d215586c6252ca2e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-26 16:25:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926298 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":14},{"id":"a0f209989406af66b96e2bf5c2975ad544fcec5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 17:42:57.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f79ee2155a254bc682127ca7eed1bd45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a54177392294639bbe6ce087e590c66 : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c2bdd4410224a35802073b2b3b42915 : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a814c4de57442b3809c8c66425ade01 : SUCCESS in 12m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97dbc4c883dc4414b1b71eed29931d27 : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f35d3d68d4444a9f391f68f20b5e62 : SUCCESS in 15m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/42294a9936d54aa49f6cc0c2f1d3c390 : SUCCESS in 57m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52177bd03ea74458811004046a8f4387 : TIMED_OUT in 3h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16fb970800934515aaf0c1b9c89c404e : SUCCESS in 1h 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdadcf1f120e44a2b5ad5c6949844e61 : SUCCESS in 1h 05m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ef16c160f81465d8d0f0531859c0191 : SUCCESS in 17m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76952fd247d143c6a1409fff14b2cc66 : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be92a663443641efaf4ac2ca93bcbe74 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2807e6a8b7684b3d8c791ce604373ccf : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e76b750b131c4c5588821a2ec85312d8 : SUCCESS in 6m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/00bbb718ea7d4e5cbdf7105a3e2335f6 : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7dbee0621fd64a0fac6d1033f376b56e : FAILURE in 36m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed96fdb1fc5045ce8b7d237a8ba4b140 : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72639d9dc1b34b35a2cadbf5f2b7190c : SUCCESS in 1h 51m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f55bcb04317f40618f22a024aa34a28e : FAILURE in 2h 42m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fee1ce17e53440739026461a1c98ae23 : SUCCESS in 1h 53m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9200b010c3174385b849ad49b19a43b0 : FAILURE in 25m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9bd97645732248da94c8fbd9d7bd1f00 : SUCCESS in 34m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3071e23064ca41f1a05c83d4fb397a37 : SUCCESS in 1h 14m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e9e0ed0bc3246b5946562dd4acdfafd : FAILURE in 1h 27m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c579f971d12b4bc1af1a5b58a6ab1851 : SUCCESS in 41m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f34ba8bd3e9d4f3d8def26558ed1bcd3 : SUCCESS in 1h 31m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/57d5eb82902a4024a02c92b408840578 : FAILURE in 21m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c9504e1a7bc41339fbd5737f09f0eb8 : SUCCESS in 1h 43m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/322afca6bf364d90b59126649d5ef26a : SUCCESS in 1h 29m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60e19180645d4bbca2a005d62cbb4480 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":14},{"id":"b4b1ced447cf8097f3e9039d92bab2a9cba560b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-26 18:09:46.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ef9c6d920ec043889f0acd2db7bbbd1b : SUCCESS in 1h 12m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/742540f9c8184888ae376d245b097088 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":14},{"id":"baf08c5b4ceca3927fd3bbb24ecc8ed1e5408d6b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-26 18:18:22.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/913/.","accounts_in_message":[],"_revision_number":14},{"id":"c06d9acdc8e28cc844e0148bc9ab754255f38c03","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-26 18:27:06.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 19m 09s","accounts_in_message":[],"_revision_number":14},{"id":"01c336fd4413774fea9ed92b5fcda0f8375134b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-26 18:45:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/787/.","accounts_in_message":[],"_revision_number":14},{"id":"999525d185ad09efd10b111f75127318bc68348b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-26 21:42:36.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"195df705289960783011c9c1a0c429c166aad71f","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-27 07:27:53.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"552f3e9069205d3ccc174056c8274880cccf68e6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-27 08:27:18.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/98dff3365fcb44ca941fba686e46ac92\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c6f0dcfa52f4a799f9c7b5ebe37759c : NODE_FAILURE Node request 200-0000054654 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df79ac6347fd46f99b83b9cdb072479b : NODE_FAILURE Node request 200-0000054655 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea4f62b9570e445d8cbf33991dd2dcf1 : FAILURE in 46m 57s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be28f55e2e6046dd9dcc7e845f5e6802 : FAILURE in 47m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bc37b11dbc344cdaa02dd94dd46025fa : FAILURE in 47m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9212a0379082437e97e735d445967def : FAILURE in 46m 11s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0667703ed7e948dd9fc628ac9fb64993 : FAILURE in 51m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/731e53fc402e4818a87153f4e8a67e0a : FAILURE in 58m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5b73e891af84fe49111a3c56460b140 : FAILURE in 51m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b8502a1a76e4337a66beac9c7e41d19 : FAILURE in 50m 20s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"d07a23f586520658c68d54eadc19abf9a1f40bc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 08:44:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e129cd19099d46d3904de472f1fc1001\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4595a45e2525425aa9b54b530b220d1f : SUCCESS in 23m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7bef1028645d4e4498deaf4a931f73df : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6f7a3d07deac4342b7bb5267454be97e : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3b25fafcf6a3eb360ad0cd5f15bcbb7acc5479e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-27 09:17:48.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5a4d592867934e5f97b818aa416714a4 : SUCCESS in 1h 39m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ba06cfb55d5f4a26a14fd55cd9599278 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":14},{"id":"331695a42b4144116bb289dffea9ab089c5c4ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 10:40:42.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72415af5eb6c41bf9a1d11cdee2ce4be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac89edd101ef40eeb27bbea0bf3a1e83 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efeb84c61a5e46a193ff1be4182b52bc : SUCCESS in 10m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc8316e9d0ff438e92bad3b39fd99430 : SUCCESS in 12m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef83c5c04acd489ea2486b31e20a8fda : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2675a76a567a401e98423ce493a1eb1f : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e1d90180c6b341c380174f8e31bb4050 : SUCCESS in 49m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63c06f1a7ea34d5390104c854dada833 : SUCCESS in 59m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22832d5abdb044ab8e594736700966c4 : TIMED_OUT in 2h 05m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd1d2d5c588840a891c5e2ee700156ae : SUCCESS in 51m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05a8be00ff744e63b51df7d9e325d279 : SUCCESS in 17m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/031a35875ef346b28cbdfeb5ac4e2779 : SUCCESS in 23m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/36b75c6f28ee4ae0b827442fa714fe9f : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7e926910bd54960b78fee0efda92e20 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/75e0bf9e6ef640cd95e3664a21e74bf1 : SUCCESS in 8m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d03c331bc944f968bab67d507c3db53 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/35b5b5ea70474d0799cde8b5d0bc5fe5 : FAILURE in 41m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b06093fc6a8140f9a9c355a179f0ed74 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d413b339e1747849507796df1809542 : SUCCESS in 2h 33m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4dfa508e255f4c93880543cc2afc8493 : FAILURE in 2h 50m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/03f687197e4e4d269ebc85e64a3812d7 : SUCCESS in 2h 29m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9e387f01641e43cca5930354d426e798 : FAILURE in 26m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/26e7e50cbb8d49cba36eb4b76c6a877c : SUCCESS in 1h 19m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/68e687cbcae64ac4a647dc1ec8347e56 : SUCCESS in 1h 00m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09c7bc930c5c4b3e9034a36314e86ba2 : SUCCESS in 2h 11m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/47f702c2a1744f838ecd8291b93391db : SUCCESS in 1h 23m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9eb71c076e754b6a9a0c426f4ccd516a : POST_FAILURE in 2h 16m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f20d6b584c2e40cb9c32a7e62434c012 : FAILURE in 24m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e63dd90a31664f5caf29a8f4420e9332 : SUCCESS in 2h 35m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da27014fe3284fd2b313fa92816039ff : TIMED_OUT in 2h 03m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04f8d1f3269b4b8ea1df9c2953fb0298 : SUCCESS in 1h 33m 30s","accounts_in_message":[],"_revision_number":14},{"id":"b5ea98a6422e362b12d49dec859dfab275cc162b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-27 15:26:52.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/14/20250527194721/NEC-V-ISCSI : SUCCESS in 4h 17m 36s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/14/20250527194721/NEC-V-FC : SUCCESS in 4h 39m 28s","accounts_in_message":[],"_revision_number":14},{"id":"d9843a836c59a6027dece10ce8916121b6c16ca5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-28 08:46:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/14/20250528113500/NEC-ISCSI : SUCCESS in 5h 44m 21s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/14/20250528113500/NEC-FC : SUCCESS in 6h 11m 21s","accounts_in_message":[],"_revision_number":14},{"id":"714ab491f828127f1217f2b03e84781792dc3efe","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-28 13:57:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d7dabf66b46cb56ee05ce9602e3aadeb5a6eda39","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-28 14:29:25.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/935ded04c22940cfb1b4cd2d757c6a0c\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bcf17c9d976f4263be495c99d584cd01 : NODE_FAILURE Node request 200-0000054783 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f029c707567e444f9b3c5316353fa6ea : NODE_FAILURE Node request 200-0000054784 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/886c001c4cb14316a7d81ed8388ed612 : FAILURE in 25m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7e0f9ae0a984eb28c56ddc6a342069e : NODE_FAILURE Node request 200-0000054786 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/793e6bb8eea147bab3f39cf974a094af : FAILURE in 26m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00769227b65c41c79b6d62e6d0423f37 : FAILURE in 25m 27s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fca18e4142c45d7bce38a9139db59cb : FAILURE in 29m 03s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c6c411b2769443aaab96430e42c9989 : FAILURE in 29m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef6f229bee554747b29f7b01f4b48ba5 : FAILURE in 31m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac605e99fbb04603ba90cb47a01b4d2f : FAILURE in 29m 18s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob dev... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"171b5f1218143edbb0188bc0a09af9dea3fe7a3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 14:47:10.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd8f41c0fc341638cca4b664bfe3c55\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cb88633c86d47b8862852e665c448de : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9d7851de7b842849d13ba77a690aee9 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/abf966b1b06d4ef79bf9a6ce66de0f39 : SUCCESS in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d0e690c9fef2acd47df8340e5910223c33b77568","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-28 16:11:58.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0d36a904e4644a8e96d46be909ee151c : SUCCESS in 1h 11m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aa134466aa7944e8ac44cd9681473a80 : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":14},{"id":"f71ebdafba99c8ae106c57ce137fa9e5a4a90728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 16:53:29.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c5354f5b25b479b91b3e2b78edefef5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ddcf433b4ea4aa9a96ecc9acc344c02 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/219c9d53c7fe476e8b1c6207ac592ac4 : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d27807e54dbd40929198dbd245afb4f5 : SUCCESS in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d4905279fae448287398e56de7af3d7 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1a2fc8e7c14a6581fe77caf0ef5afe : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/66882d7b8e864c23bf4c0ab82e3dffae : SUCCESS in 54m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/282d5c1656824db8a627ee3091ed074b : SUCCESS in 57m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccec0374453942caa48b1dccccbb2754 : SUCCESS in 1h 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98233a932dca4b83a20e617778e1ec48 : SUCCESS in 57m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a178be899559406982116535070782ee : SUCCESS in 23m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/75894a5518874f268fe7d6ffe349eecd : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ecab5f3f6984b77ad71cb3a85889c7a : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1a264c2c8a84571bd2bc5bf8c2aeb43 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bbc1f2e2e0104df0900428f3fdcd1294 : SUCCESS in 6m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/78be07e49bfe4fd29fd1f0051cc0c706 : SUCCESS in 8m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2ab87195ac9143cb8a9bc6b222bd936d : FAILURE in 35m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b980294d4756431cba939488697dda8b : SUCCESS in 8m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04554b3f059643eaaf1fdf585868b689 : SUCCESS in 2h 18m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb4f0a83591d49d08b77a8a71e3cd1b3 : FAILURE in 2h 45m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73ef9730e68d4dbea86bb7bf563d6852 : SUCCESS in 2h 09m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ad9e969caea94a64adb644d36e8069c8 : FAILURE in 23m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/179e44ac84d247bcb89b022ff2990e7a : SUCCESS in 32m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c81777f08d48443c8daaf7faa8f2224c : SUCCESS in 1h 08m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67ad74130ddf4a45ae50361037a74de2 : SUCCESS in 1h 53m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38052c977ff04d91a048311593e5d1e6 : SUCCESS in 44m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6de2cc1c961a4b0cbb0048e73dc09602 : SUCCESS in 1h 26m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/41248614978b4d90b8e548e1df786d24 : FAILURE in 24m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd30aa73f87b449e8651a576d016c074 : SUCCESS in 1h 34m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cf986dbf43649d1ae7d448f31cc5c0e : SUCCESS in 1h 26m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9cbba7375ad4c66b51c6d543441ec2e : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":14},{"id":"4b5c166d0b45a2466fe49a1167e24d514c4d82a0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-28 19:55:08.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"3a4d7c41e2d277881afb540c667f08814dedee42","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-06-02 14:00:36.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"62a8b8914b5301203fcb86d124df6d08516ddcea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:39:43.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9785ec1aca45068d9090a35fb931c6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/feaa878d73be4c34b772a63a80e40aa8 : FAILURE in 13m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f06c557140d41f9a290e5525973fde7 : SUCCESS in 19m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d72b4a83e694c97bcaef384fe99c088 : SUCCESS in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d85c1b42b411d30aa383a5f32d4dd3215c67a360","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-02 15:45:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7ee9c9b1ef534f7fac5f23613b4439ed\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42a869124aa048168cbc1fa3555f0cce : NODE_FAILURE Node request 200-0000057228 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfcaa6fbffc249eead8761445c9fd3fa : NODE_FAILURE Node request 200-0000057229 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f6f718126484d83bd2805461ce1591b : SUCCESS in 47m 29s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43f1531abdcf4143ae2b0929ac7a1b06 : SUCCESS in 44m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1dcb8d4a5844aeda4e389f68f3bcb88 : FAILURE in 1h 39m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c0a282c511940798c9962b0980ece61 : SUCCESS in 46m 38s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab1f8b753c59455496587355cc0e9885 : SUCCESS in 52m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d9b7ee91239542b7a1f3da1c68f5ff11 : SUCCESS in 54m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c058965983d4901b2d7aea11c1c9bcf : FAILURE in 1h 44m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b21a050636ca477497a80adf881ef0d8 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":14},{"id":"745c745f4cceddc0e55b97900e2aabf138ad5784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 16:36:27.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bdeb3afb504a49b21d4128d92b9842\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbec5c0cc104a588b58f33ec50e5853 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c941ad9e00034987ac30e8fc6db363a1 : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32e29e1f14c64c16a1b4098a309a1c40 : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37b1fa0cbc8f4832be775fe58ba0c0a0 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed860cd79404d2eabb69be67ef9b3fa : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/903e86cac5864edfa62f0f49506e946e : SUCCESS in 1h 01m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/674663374367457994f7d32f95357f3f : SUCCESS in 54m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d554fd49393a4a84a3e7ca392081eacd : SUCCESS in 1h 30m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a984cc927304101837ba71a3837b5df : SUCCESS in 57m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5610565ca0cc4c54b0ea573cfff1372f : SUCCESS in 19m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8dfd7fe5acc349928a110d7ebd2af480 : SUCCESS in 24m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ada01811eab4cff834aeda6e6c7bf7e : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/570cdf2f4d9f447d983f9df249563760 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b4ddbea51b04719ae9ada880a502951 : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/93c4661753174e13ba9066653cc36899 : SUCCESS in 10m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dad32dda3b8b42ab8370e14dbadd3a22 : FAILURE in 36m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59728ab8e229431394271adf5efbaef0 : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1208b40e5da431489d395645e3ef059 : SUCCESS in 1h 46m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3a823af7835a4c1690eafaad077a0724 : SUCCESS in 2h 33m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/854864a119d54c24937794b05d830cef : SUCCESS in 1h 37m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/739cde36380a4e58a163c475e815df88 : FAILURE in 17m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/902e12b7d4a740a2a461b148367386f5 : SUCCESS in 30m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6cf64e69ee8c4e7f859a0b3c59ec0bdb : SUCCESS in 59m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8a4443d23234aeeb0b188cc89cfc361 : SUCCESS in 1h 25m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/389c905648d94ddc912d85206ced655b : SUCCESS in 32m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9494462efb8e46b6b41f4480dc429096 : SUCCESS in 1h 18m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/281a94637be743718468f766a6010f68 : FAILURE in 18m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3505ae89c684479848269deba287f74 : SUCCESS in 1h 28m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0350f81c15874be3b1252c77b281ee32 : SUCCESS in 1h 31m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b16ab96efa5461cb7fb16d1970fab3e : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":14},{"id":"6d6bb20983d7633e49b76d9ec9cd5a2ea6fb0c29","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-06-02 17:41:20.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"55b1276ffac029203e3f88e9444a10af34963952","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-02 20:59:41.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8ba1d55f46474890abe0c70ca20bc53a : SUCCESS in 1h 28m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/45d89d7ca6c44cf287ab58ef96a43264 : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":14},{"id":"2c9a1615116c4df64dbe7cbd9b291ab547f6a77c","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-06-04 08:12:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2fbb1b5dd317746866eca5e0f30a3db66cfad8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 11:54:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee4581c848224fb181cf0a8a5602f775\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f593e9f60c4fdfafc68ec79a179b1f : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c7b00c56bfa41b9b95fd19a442d13db : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4897765105e14f1a965249e17bcd690c : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e38e41548e414da486b0cdb31fe88582 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb60410ee23345a29dbc768d29c2e15a : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/90e45da4a62f40edb366fae9d60fedf0 : SUCCESS in 54m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d40b68564e3474b95019cddcdb0b9dd : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ea15fe80baf442fa8a3a55ba53c1249 : SUCCESS in 1h 45m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b61e6588315d49c98877869ec353929a : SUCCESS in 1h 03m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adc3a01d12354d84b4c13e96bdcde577 : SUCCESS in 20m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32d393d2365a415f9666560d129d4164 : SUCCESS in 33m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8284d056763747468ac167bcc44ae68d : SUCCESS in 6m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f5005a8b7994914a7a8138d9452d563 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e78b62f392de47dcb6df6cb8564c5d4c : SUCCESS in 7m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e0dd58f1193d4bc7920cd2855dcf0db4 : SUCCESS in 9m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/490beef471e547b9a54b9adb30d16a10 : FAILURE in 35m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3dae7ba580e54284a89edb2f687b253e : SUCCESS in 7m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04463250f73444389157046ef48e0a06 : SUCCESS in 2h 10m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3f0421ada824e2d84b71a869d117e84 : FAILURE in 2h 22m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1a8746919b2c42679ed7ccb7d59767fc : SUCCESS in 3h 05m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/09c66c4c2811496db172b73b8f0a3b3b : FAILURE in 22m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a436e5de43db425eaf54ee1aeff2fb25 : SUCCESS in 32m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e941a9b397054012842c7f246789b5aa : SUCCESS in 1h 05m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/358b3c76e562439795d202ae80b2e0ad : SUCCESS in 1h 53m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c705b6c775fd4b69be0c6971fd766580 : SUCCESS in 42m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2141f7f3b5e247b985c20d122540a0e1 : SUCCESS in 1h 33m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43150a10bc8d4bbb91603ce73569a8b4 : FAILURE in 22m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05f9750999aa45e197ec9d1a7136aa48 : SUCCESS in 1h 36m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72492c35517c4aaf9d0b9ef42055406d : FAILURE in 1h 14m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e34a1c150c343d8bdeb6cab60ac7a09 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":14},{"id":"9572878addfcb3df0de8666a0452c62e0b848a27","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-04 12:16:51.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1711fb0a49884ed3b447d3d1bc27422e\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9578e8d58804a148f620c42c01a46e4 : NODE_FAILURE Node request 200-0000059093 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71a1b5f7b1fe427a8f0c52df8d76eed1 : NODE_FAILURE Node request 200-0000059094 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/92b5f3c6df884971bb531af2c59a9933 : SUCCESS in 45m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7fc62d96ac454d88a0aabfb1c2f0e767 : NODE_FAILURE Node request 200-0000059096 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d30a7f2db53b4cd396bc8457356b5635 : FAILURE in 2h 05m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4121f7c8ad8a4544b6f0bd0fa60ea281 : SUCCESS in 46m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17b4e16c39f8423b9b5aa93e120d1b8e : RETRY_LIMIT in 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3bb07ef15c5e49c79e477df576d3a5b3 : NODE_FAILURE Node request 200-0000059108 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b58f78badea44cadb440646cb71e8564 : FAILURE in 2h 20m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e08c275ac34f4320a248315c19dc02a6 : FAILURE in 1h 04m 14s","accounts_in_message":[],"_revision_number":14},{"id":"9ee5b625d6c186c5bc68c1e2f58156fcee250952","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-04 16:10:44.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4dfc1c98534d4470a31672fe0d33fb4c : SUCCESS in 1h 24m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0f30a432c74a4725bb38ef6e9bcf76c5 : FAILURE in 1h 24m 25s","accounts_in_message":[],"_revision_number":14},{"id":"a8418c176f4c307c797633cfd1832267dca44321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 17:02:09.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa7d5057c86948edb2b081de8c0a1995\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6dee5d0ff8f45efbbcd7de170879bd7 : FAILURE in 41m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aac521d989f94a5a96ac5780b8f25c58 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/022a4cb94c76442396c0112dd4b9d236 : FAILURE in 52m 21s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b58662250f88e3febd055626aa74cde163f37430","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-06-08 14:03:52.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/926298-14/Seagate-iSCSI/ : SUCCESS in 2:25:21\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/926298-14/Seagate-FC/ : SUCCESS in 2:12:18","accounts_in_message":[],"_revision_number":14},{"id":"22a3bec80e6e766dfbead5e24c6afeb866a7174b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-06-08 19:04:09.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/926298-14/PVME-iSCSI/ : SUCCESS in 2:25:01\n- PVME-FC https://os-logs.seagate.com/dell-ci/926298-14/PVME-FC/ : SUCCESS in 2:13:21","accounts_in_message":[],"_revision_number":14},{"id":"13f0940e413b932edcd28bbbd991e171cca0dc53","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-06-08 19:05:04.000000000","message":"Patch Set 14:\n\nBuild failed.","accounts_in_message":[],"_revision_number":14},{"id":"637c22d26f8ebac70955dd215637a572eeffef6f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-06-09 00:05:44.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/926298-14/Lenovo-iSCSI/ : SUCCESS in 2:25:47\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/926298-14/Lenovo-FC/ : SUCCESS in 2:12:42","accounts_in_message":[],"_revision_number":14},{"id":"4f145280ae90fdcf79248d0852474c0eea719154","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-06-11 14:34:12.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e8b2a9f4091cd4583eb6a2f61f7fb4e0f1d84575","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 16:11:20.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d4fb2196214445988b8ef341b38a17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e2cb476f467429687cc0c413a18688d : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a801a2d428064ae9b5b6bdb85172326e : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f059751b3e746bcb27538334dc35eea : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"343f9ddf86f4d677c017291a54c5d477e3fe06df","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-11 16:26:21.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fe87f725909e4d12a769bf200fe42308\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e9f1c4ef306440caa851f9b5cdba656 : NODE_FAILURE Node request 200-0000066920 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01c4d5f46b9a43d9ba6c1ff316cb9e44 : NODE_FAILURE Node request 200-0000066921 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0780a97579474866998e72ae97432355 : SUCCESS in 49m 32s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6550526af3f240578223765e6146f398 : SUCCESS in 47m 49s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/deb0146696ce4328b18c1a5064cf68b1 : SUCCESS in 45m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bab644277ea4403b837fd90c028463c6 : FAILURE in 1h 03m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9dd523f2c13e4022bffb8907b8e77546 : FAILURE in 1h 51m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dff9044f234d46f78ba99a806b0ba31e : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":14},{"id":"68dbd059c801159a7d7d6aa16ac4ebefcf527df3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-11 17:10:28.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a51498eb6e65493185f1857b9ae4680d : FAILURE in 1h 19m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f6fb8752fcd34d89b45cdf700bf9ea0f : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":14},{"id":"8014820f3e1f4b434b6e0de9c0c4290c60a42d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 17:45:32.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3535e55c59284ee29c15ece44ab60f97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6875dd42faf140ddad73c8badc309bb9 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e71f5e0e510431d897d4b7893164caa : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59f11d63ac984428a545babe11d0548b : SUCCESS in 13m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76883d3481b54b6598e470b306b52818 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87aa8cacda34805a83cff938a95f9ee : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9482fb6cb53a4b27b443fa19efddede1 : SUCCESS in 56m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea79ad0e5c1f4851b21b190c19d79fb5 : TIMED_OUT in 3h 04m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/227fd846ad864a1184a0952ae32174ce : SUCCESS in 1h 44m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b003edb734584d539a73094662642e39 : SUCCESS in 59m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cc3fd35e345481caea1ba9d83b7ee29 : SUCCESS in 19m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7678c5486ad343e0b5a97249745bb919 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a092522edcfc4f49a507244514e62464 : SUCCESS in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fe3cc284283495cb099fbd4f3f2d3c1 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fcda218c6264ac8a4c0b57d07669178 : SUCCESS in 6m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92bdb9e9daff41ebbcc14440b85e1d60 : SUCCESS in 11m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e95fb22482dd44dca33125acfe190e7b : FAILURE in 20m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f085ca2369a44fdb045d2b00b87f059 : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c6de15b479a4fc3a7644badebb6f3cd : SUCCESS in 1h 52m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f24437b18b8c42dd97579250fe0a169c : FAILURE in 1h 19m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b6e3d3068a04cbc99def88383d561f3 : SUCCESS in 1h 42m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4b6bddbe9b3a4b25b01d219ccfbe52ee : FAILURE in 19m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4cfee3a319b94902bdfdfa9ffd5a6ef7 : SUCCESS in 32m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c029c8548974c389da676f7b4a3a53f : SUCCESS in 1h 07m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b60f8ad19bc64a1793cfe9e5eaf47562 : SUCCESS in 1h 40m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/50d6fc5090cb4debbbcfaf5823d83e72 : SUCCESS in 36m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3eb7e409455146bcb223abc1e97e4771 : SUCCESS in 1h 37m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/98c22f1102c2485187af9de75c756967 : FAILURE in 13m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82fdfb45d0cb47d39dc4419486d44d15 : SUCCESS in 1h 35m 33s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eebd1bdf25494308b16827f00df67812 : SUCCESS in 1h 22m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab9eae8a9a64f26b0b57c34106b0047 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":14},{"id":"2d0a04d73778c9c3f13b1a6e0ae337559892fd6a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-06-12 14:48:59.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c1aa10b558aa5173b42c71b9d17fd401d12017fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:38:17.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bfd4343b7b449128d771d20354bf936\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d13edd1570446aa92d16104a2e443e3 : SUCCESS in 15m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6483c0e40dbc43d2af282dd6e0328b63 : SUCCESS in 29m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2cc7b4c60004548ba3d15fcdae69f1b : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"10a74b52ef1621ef265e70149d40989759fd83d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 17:34:30.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39e7ca4b43ff4635b28312cfa6ad3a48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f263d6e1a74ead8e5c584bb6794b45 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bccdcc635714fab8442e2a3ed4881c4 : SUCCESS in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69b3765e35714f35a06e6305c8f3410b : SUCCESS in 10m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de5ba15047a34e5b95dcdd86546bdcda : SUCCESS in 14m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d3d4bbe1ff41ce82b8194e86e4e14b : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c92962e5f1364bd9a7b7e705f2a6841b : SUCCESS in 1h 00m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59bcf7ac44394816a9a896154740e19b : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/033a857bc5884057ae33a65162b4c20a : SUCCESS in 1h 43m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5a98a4cb7384e9c961bf7736a6d97ed : SUCCESS in 31m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7a2ec88748a4c24a2016db534825b0b : SUCCESS in 17m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb337fcd1e614a68b39c2a424d3e73af : SUCCESS in 17m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f3b4da38bd57498894abde400cfa22e2 : SUCCESS in 7m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30792de1e7464f0eb613831438a1d17d : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad4d5e50fcc1416c94211d16b1b5f2b4 : SUCCESS in 7m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e7f6906ffa948b3a51fdaffdd381396 : SUCCESS in 9m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8bed35091c647d68f8affffe1ae13ea : FAILURE in 33m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d763f6ba9874530862ce859de55092f : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a9b24eea47944909aeccb6cc47622b8 : SUCCESS in 1h 55m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b8be8e66b6d4d45a904facc38581b6d : SUCCESS in 2h 33m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0ea67c31604e4ce5ab968abf2c9ea3fa : SUCCESS in 1h 39m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d17df079453443c5bd327f0a3e3a305a : FAILURE in 20m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04e6f34243a94776ac42469ab1cf2172 : SUCCESS in 28m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/556e778168374074a30ee063e4da7772 : SUCCESS in 1h 10m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0aa78d2ab23a4535aa7a6f196a94bf71 : SUCCESS in 1h 31m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24e5830bca7141d1bffd2154a950884e : SUCCESS in 21m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5023e16efee54accb8444fa4b435c22f : POST_FAILURE in 1h 58m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cf1232253f13447694e900536ca9d697 : FAILURE in 12m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb953b154cc846f68f6f8ec6beabb147 : SUCCESS in 1h 41m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7970726d5dde4f929d6171e4d7589f4d : SUCCESS in 1h 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b0bf41708a248b5be76d95af0684c20 : SUCCESS in 49m 52s","accounts_in_message":[],"_revision_number":14},{"id":"1b807ecf7599017c5f55775a034b876db99c2549","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-13 02:45:19.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e7fbf3be0d27480396d0a9c5d7abc95f : SUCCESS in 1h 14m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9a00419b4d8f4a85bf069854bd16fc18 : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":14},{"id":"a0de1d246eccff504063437c6260710a91d42a13","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-06-16 13:36:21.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"cb6987d6759ab14d7829ad3e3e4a8f7c5489a375","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-06-19 07:29:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7d5b4639d27950cd6ede462320c5edecf4d1ffae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-19 07:30:54.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b5d57f1bdfca4e7fa431db741cab45e8 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6f54f033cfbb42bb84724806a8bab723 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"b0673428af87a5a2e87fb4b43f2f338f3793667d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 08:11:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9147d1300c4b4145bc0583f2750ac87a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ca90b73db364e1f85e7cff27f96b988 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8c123f4bdf441b99eb259cadb36a1e3 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42cabcae865a468581d9c6d7c54a21df : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"594f980df00153529a8095013a31655b0c0c4746","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-19 09:20:33.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/67150153a778415ead3ab1ceb925fbc7\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4130f617685a4e008eebf3ea5f5ac13c : NODE_FAILURE Node request 200-0000068433 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6c2e30308104786a333b8e8993c2c75 : NODE_FAILURE Node request 200-0000068434 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/721d980996464d1287b9d5f0ef88fceb : FAILURE in 48m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95266d94319b4a6ab234266ae295f122 : FAILURE in 1h 50m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c31fe7ff3dd464da25457f145fb0c35 : FAILURE in 1h 02m 25s","accounts_in_message":[],"_revision_number":14},{"id":"bf1dae306927c4236924b6b8f033f1e0c5ceb0d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:47:36.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e7606d352cb440c92dfa59544013e5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d3db79b5f8435489962236cd4c3a24 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16ac5880c58e47f4882b9af29e784d11 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/691347cc3b5d4d59825f7bb9d1d72597 : SUCCESS in 9m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17ded227be044ce483ea9ca3d2f9cdad : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd0a37d0a8547cfbef444474b2821c4 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/748b4758ac444e0cbda507e23e292ecf : SUCCESS in 58m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5894894505f94c91a32fac1c878d9f18 : SUCCESS in 56m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2e0b81d6fb947b4b0b8c5e3e7a7fc92 : SUCCESS in 1h 18m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83e5f8dac1154029b082c5fd31d6bdf1 : SUCCESS in 54m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e17d50db97814acd8d4cc6726599ee97 : SUCCESS in 19m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/64ef2ad6d92e43c69bfe2aedb4ed6884 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e830e73ee8f40a8a10f9748e9e7d74c : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/294387c734354548ba9c0533651a8c54 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/469360cd3e5b48a5a63ea72ca3c15f98 : SUCCESS in 6m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f693514017f844d0a32e01adc0ceaef4 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b027c5674a2f44b29d98ac1d2285540d : FAILURE in 33m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/714463743f5f4fe6beb17a02ed4000d4 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ae11253a66645a2b3128c99feead997 : SUCCESS in 1h 45m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41b6b7472174478b842cbe7f7aaaac9f : FAILURE in 2h 16m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ca836161047a43378c84e7a9649a0a0e : SUCCESS in 1h 42m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1c197505060b47c2b326f0e6b5e5f0ed : FAILURE in 12m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f50e40542c784e448a406fa7fc467a93 : SUCCESS in 35m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7ff40da68cc42c78f8ddfad3c58bf43 : SUCCESS in 57m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61a1f0ec6cad4032be4e7e24f6a88001 : SUCCESS in 1h 40m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b4078be45fc4fedbf95049e5f54795e : SUCCESS in 41m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ca4a930cd46d42908fb44842732035d3 : SUCCESS in 1h 30m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0ca4cfb29324dddafe1ee6368e875cf : FAILURE in 19m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07ef189d53b041fea0ae99d6d8ead52e : SUCCESS in 1h 58m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a09449bfa354b6c940591b80a00a2f0 : FAILURE in 1h 44m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45d3eb470d454cd986e164a5f64ded23 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":14},{"id":"3f568e847b35bcd1c898918f6318005b3b5c6d22","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-06-19 12:25:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1d213ae0b5e0b0d74e5802fb16ff10850806a3b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-19 12:25:59.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9aa2e08efc4546298ebad963a6486d76 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cc0dc96552234f978e52d6b63132db60 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"c390ebe26f055037a182693486ef8eb137a01bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 13:28:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f38bdd30b4f496f97d54127c3bb1118\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21534310bf2649c68b05ee8438ba3bb8 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0c13eee9d8f4fb5bb4a5eb6886e15c7 : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/554dff46c8da4a11a1508b13b322d9ef : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"971f6104522744c62ba436e6fa2f680b32c6d926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 15:28:18.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1191db7ef7d84c2095c017f67e674817\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688e62e22a9444b5b7cd7289e5dd4bf9 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26cf724c40ac453e8d13836ba83e0694 : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74e2ec7347534f539ba7a0e9d5c46d3c : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c1de65bb79f44a2bc8e7667c55fb248 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acc742b4a3548799622c04801e22549 : SUCCESS in 15m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/2edfe9590cf440cb9be1935ada4a521d : SUCCESS in 55m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20ad3bd06b5a4122a62c0ffed68f5ebe : TIMED_OUT in 3h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de7d1c5263bb47cc96df444473fa9d28 : SUCCESS in 1h 40m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a88013b92e4e45da876a2da45255a16a : SUCCESS in 1h 07m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56974c0dd6324444ba32b8c567a1fd30 : SUCCESS in 21m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/451c69531b924f5c8c8373329d540a98 : SUCCESS in 25m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b74310cc3cd42f6a003b14e79da9f36 : SUCCESS in 7m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b3c4232eb4c4c309baf940b3f73824b : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14902b723044446e89eecb7c141aac4c : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd9cdb6cb953471eaad246cec3b228c5 : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c203134b81f74bbd847e234024b2ab29 : FAILURE in 26m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6231e46cbaa641e6bf06b412770f753e : SUCCESS in 8m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e22387fda8e0423684c17611590591a4 : SUCCESS in 1h 42m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dba8be2fb67f45ea932b3a51012e238e : FAILURE in 2h 45m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5396ad72cddc4df7acb6ea8e2b7f03b9 : SUCCESS in 1h 47m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d67a8cc3bf34446b99b5de7183d61b8c : FAILURE in 14m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/02c7dc3252a0412b99cddbbd168c5065 : SUCCESS in 26m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ba8a9d162bb478386f40126e00f0980 : SUCCESS in 1h 08m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/597062a262274246a4693d3fdb9bdce5 : FAILURE in 1h 33m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/209e0ef61e9144cc9c1a03ed87012e0e : SUCCESS in 38m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8c0b7116ce342ef806b2593acc2b620 : SUCCESS in 1h 14m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43025616deef49a9803f47161b8240e5 : FAILURE in 17m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/800800564cbb4bb99468224371b148e4 : SUCCESS in 1h 35m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c07e9a7ceba48909452b3bc60c00dd2 : SUCCESS in 1h 08m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21219325a0374c1f8eca9ab4d7e965d1 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":14},{"id":"c9d25ca4ab67463adae618f84499811f838aa988","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-06-27 09:19:12.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"822a1fb7382fdd93e399d9b706bb3cd5c7ebb146","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-27 09:20:12.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bc109911389348639ba4cd3988ee5216 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0c78f335227b40d090dd58c7e0291485 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"32944903108de2ce72297cf6ee4e710ddab41a36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:48:15.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1ef7dcea24484093caddf1ff258be8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6124d4dbfdc4c3190a1c1da99050c58 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb04cec1a5df41f5b8be02ef336eef4c : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f98bec315ac44eaf98c136d38e7fd0d4 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1113a71e5b67ea19f753a5ffc3c0c30153007dcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 11:20:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50818a593ceb4bcfaaf111dfb260d7b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c1e5ff475b41379774baca9d984734 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72ea0c38b8a44763b72075210bd351c2 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c55796c209cc4cbdaaa24c2ba72f77c1 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/363ea24032264d9ca1a702d4184f376c : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea5efd372864ed5a471566a98e34aa4 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/043ed19560494499b534ec82c07bc466 : SUCCESS in 50m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb25d9be714e4e4f95bf16a62e9c9bf3 : SUCCESS in 1h 00m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee4c44438a3240e28b0124d6290b8418 : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37b549f0770441f2a44ea40bb2876e59 : SUCCESS in 59m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd7d1301e4024f58b4bbfa75b74fd9bf : SUCCESS in 18m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4eec118a08ef43598472985c800b16ec : SUCCESS in 24m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf4429dd0c3746e3bfefbaea6aa00a9f : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b754f7afe7364e75adf5d51381b1dc26 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0935a287f872495ab9c04870fb96278f : SUCCESS in 6m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a39400f5a82d4e64b24c5709fdb5c86f : SUCCESS in 9m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c80e10b7cd34433bbb9fb2e9ea325bf0 : FAILURE in 34m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2de3a7888e56473b81618384e034c79c : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f2b64438c614a849203f1e7eb576dd2 : SUCCESS in 1h 59m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/048cd48b208f4f1fbdb08d955fe7f795 : SUCCESS in 1h 51m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f2fdf04966fa4757ad7288c91c25fb89 : SUCCESS in 1h 40m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ee32c57fbd8c4006a6f370527952f471 : FAILURE in 18m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/40db6879854d4e09a9e86c4a08e8bb99 : SUCCESS in 16m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d35079006124982883b1afcf08cc4ab : SUCCESS in 1h 01m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dc1e65ce8cb14b1fa26fc213de835d32 : SUCCESS in 1h 29m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a337e9c726b942da9cc8e8a89848bdaf : SUCCESS in 41m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/478f85699fef4079a049f650fafbb30e : SUCCESS in 45m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9fe584f70f154ff9b728dc22efceac3c : FAILURE in 19m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/090a71998c1f4203836194ba6ec68e65 : SUCCESS in 1h 43m 31s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2927e996b9449b0a39ca92fb1003649 : SUCCESS in 1h 34m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7b95237533d4a40928d6661ccd00277 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":14},{"id":"5320aa060b854ab46fa2243735338a493ca01721","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-27 13:24:45.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/83097c9e1873480d851e4042d0e7f500 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f5b61b4a82ba4e609c20e5e65cf6f2cd : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"f30afed000147b9edda654a69841b9f2c6bd9c01","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-27 13:33:52.000000000","message":"Patch Set 14:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b5720afbd81f300e05f844d883581fd7aeffa572","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-07-07 20:05:00.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d7936e4c3b442621870e1c1998544d9442e9666b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-22 01:42:50.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6862c69ce720b72de08d09276629aa1f05990a49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-18 14:51:22.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"6abc98cd1a83b05f09b57f61aa3195a999cda3d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 14:51:37.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"b41085ceeaf5d363136f0df4907684d26daa06e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 14:51:45.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"dffcbf722931b252286e46a1cacac5608d62b690","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-18 14:51:46.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2629/.","accounts_in_message":[],"_revision_number":15},{"id":"2c8418e361849e22da7fa17577ce0cf9acb4dbd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-18 14:52:50.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"746f6bee9b93ce48d4a60699b55e4a30560e1d15","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-18 15:47:23.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/523d2efa7b474907b73e2bf3b0708279 : SUCCESS in 47m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b34481a9311441b8819f6b1788b79900 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":15},{"id":"b104057dfb2c1874bf61456d19a73fa2230bb706","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-18 15:48:47.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":15},{"id":"fd552df7090066a5a6f4ad283fec0951576bc305","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-18 15:57:11.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"539d431ad5e70924a2cb350f78ae304f8d61d797","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-18 16:25:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0c0c4e16aa3e70dd6978908389834d7e46df31c5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:17:33.000000000","message":"Patch Set 15:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2507/.","accounts_in_message":[],"_revision_number":15},{"id":"2c045a2de18f2b373d27ce2ff2ef45cdfae2a4aa","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-25 15:22:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dea54978b8908914983e44f655a47d69b788ccb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 16:13:21.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab53104fef64646bec93d6b10b16b47\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52b1728b348a46e0b82e9333d3da14c5 : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d8b47bb5959453692884c12db346a86 : SUCCESS in 40m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2eb795758a04c9da62c592fecb56fb3 : SUCCESS in 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"cb4b2288981c300d68d304f9a1a35308379e14cd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-25 17:09:00.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ffc9a0ed579343908ee00fbe46998735 : SUCCESS in 1h 18m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/40df575c8af94e2bb2cf9baefedd7b4e : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":15},{"id":"78460bd2259ae7f8f0c49afb179fe2813e12c46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:54:46.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/126ac0d853e546629a8f62b525067a9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9559378e95104b64b5fb9fd44713fb82 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/303e4eb110934f25b528fabfd5f5a203 : SUCCESS in 13m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/076e610a04484f73ada05ac0b02ab193 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c56d3068e72470bb55f52a04ef5ebc0 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/02db39211acd4717860ba09a8bb29200 : SUCCESS in 41m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c43ca0f2416a4a409000530179620fe6 : SUCCESS in 40m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8bea8a6e6e9141bea866a83b922dbbff : SUCCESS in 1h 43m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/479453a766144cbd84297fc8c8862598 : SUCCESS in 1h 58m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e31dddfd2a894132ab223a8139f1041a : SUCCESS in 19m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d04b0d502f0349df8aa05b1f741d3586 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/602c2f33f80f4ff89f840692fc980435 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41fb2fa5aa2e4299bb641a2d4c209de5 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77df456fae744a6e9b889251b462b072 : SUCCESS in 6m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f2503b28cee647cfab15bc21ebabd3e4 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3f16cd3e6e5a4a1aaed20dd4858d9038 : FAILURE in 18m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38543455bd3c40438387aaf5388b0aa5 : SUCCESS in 3m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e3c951df0374824ad7e4a8f848dd40c : SUCCESS in 1h 31m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01b379b9932049f29e2f04deb4c78aa6 : FAILURE in 2h 25m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f094354cba2d4713943aff75f0e10fdb : SUCCESS in 1h 44m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/56d5a6af05c94722b0b2f8c0eedde596 : SUCCESS in 2h 06m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2a45fcbd2e5d419cadf667f838bbc5d3 : SUCCESS in 18m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d1d5e388450a4b378d969159a0d7f712 : SUCCESS in 1h 11m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/54821f3a7e6846abb3abe2b7c151c653 : SUCCESS in 1h 26m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc8528baf8f345188174257052658a6c : SUCCESS in 43m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0cd9e52dc5a5486fbb46a8da74939f30 : FAILURE in 1h 25m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6724a2db93e64623960854df08fb5395 : SUCCESS in 1h 58m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/551980eff6974db19a8fc7f27a5c1338 : SUCCESS in 1h 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f1419ed71c7420dbf649eaa123777a1 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":15},{"id":"8896488eb3d6e2f5cd63feb95c6df3290bb30e50","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-27 09:07:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e9f62a3d9cf604daeae34d39b058dbc86c7a26c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-27 10:10:05.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ecd1dc98aa5147a49be5a4c649c5f586 : SUCCESS in 50m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8f9327efdcc843139a58bcbfd5787442 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":15},{"id":"f266ea47595cbccb3a12906ea42cdb63ed908783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 10:25:48.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2865310b062d4f45a91ae26ac5bd8ee4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e98b4d7f0b524bb18a54333aa06a83b7 : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ffc004533e7450cae94e9f176b66064 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d766aeb38ff84d64953e7b20caa51332 : FAILURE in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"79971c0b605b0074d5917906364b4dbd335a73a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 11:27:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3646091ff51544619b5e161186671541\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a9a99cc3ef4468b8d72ec672981d3b : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee91f7236a3c486993dfda08be0458de : SUCCESS in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6c75ec183f846b3a187666133fae77d : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a4a107ad4a465e8deb15bdb9e7e649 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8ff39a4a47064ac3904b1104eeb191fa : SUCCESS in 1h 01m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b13a4ddad544bcbb287b2f1744b2bf7 : SUCCESS in 32m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3dfde6ab9d174dc4ad5277d507e417f6 : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d07e2b1756864fe6abbeccc39248c019 : SUCCESS in 1h 09m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f417c744e99d497c99b9b2e9f5130a49 : SUCCESS in 16m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22bf7c3b3c524f51907af04e515f9eb4 : SUCCESS in 21m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/17ef8eb4078b42488e92a5cf09fad4a1 : SUCCESS in 3m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a0b5791075042ed87ffac018b8f4ada : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f2445e24d3134bcd999c30a3df738c8c : SUCCESS in 10m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cd6c3bfd1aa54ab39ee63a1510c4e1ce : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11ebc376c0584f49ab9d1a9f9c940b7b : FAILURE in 37m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e7cc601e7054def9358969d81723494 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/606983214f4d4814853f5d9e528ad004 : SUCCESS in 2h 03m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b750cac84ba34d7db9216cde95067441 : FAILURE in 1h 31m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4978830f164465c9af35af48602e244 : SUCCESS in 1h 39m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/05a6f6d56c1748978e9cf682fa297e91 : SUCCESS in 1h 47m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1b208a89c5a4a5cab943e4ee6c79dc2 : SUCCESS in 19m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de622872259c473c80209d207dc0a3e6 : SUCCESS in 1h 02m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/512c38c1d4444d7d9d2052ae12481d7d : SUCCESS in 1h 17m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cd379e28c76642fc888461330c7e8b29 : SUCCESS in 48m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f5c219b1d3de4794bf717e874934c12f : SUCCESS in 1h 47m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/694a96fbd7d948588b35ce6ca7bad6c7 : SUCCESS in 2h 18m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f96df2e8da45490f9c5549a445e304b4 : SUCCESS in 1h 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9117579ed9724fea90258907d6fc8acc : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":15},{"id":"f28c900208df01730a408f1c6f5c42a4861c195b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-27 15:26:01.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e1d2db92720b1e0fca3b597236769ae420ebd853","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-27 15:26:09.000000000","message":"Patch Set 16:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/77/ : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"c1be3b120f60a1f8e283887e9eca3dca4d9706be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 16:01:40.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d6bf17c7234be09cc43c984ddca267\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7b8ef8ac4a5434a8c72971bd872dcd7 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8ecd139baec460f88808dc5a161018f : SUCCESS in 24m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eee5dd25623a47f4b1ca5a328b391b66 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8bd6b67e916d7ad1b0410f37355e3286406c1a1f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-27 16:17:48.000000000","message":"Patch Set 16:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2583/.","accounts_in_message":[],"_revision_number":16},{"id":"59dd2d2899d8f73d5244cb163bbd32be5517b659","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-27 16:18:50.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2731/.","accounts_in_message":[],"_revision_number":16},{"id":"5db5431a4495aae76f5c09f59e0a000978c32a85","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-27 16:18:51.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0edd909e752d4293bd9f913e75c70362 : SUCCESS in 49m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/db7a35f35b834da2ac832befb6b5c945 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":16},{"id":"5e45af508ee5ffcf5f15cbeeb481c62a46b9baeb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-27 16:29:51.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":16},{"id":"34845a960312853e7b360ababda5a3bd974ee8d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-27 16:39:40.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 10m 40s","accounts_in_message":[],"_revision_number":16},{"id":"759bf80068e7d4353d38aec35eac4c46f77803eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 18:16:55.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6173fda274444b4b26ff859c07fa8e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff31661512f2449dac0170902daa5b94 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e487b3baf784f5a949298878acc204e : SUCCESS in 7m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93232a387b8a4d90abc58d2ac070388b : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d458b775ade4bc796c9a5af6f6de141 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f5ee211e19214e00a24b542685706908 : SUCCESS in 59m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3dbd25abc713410995d071a0dd98843f : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2289157d809046dfb4eeaa54901c20c9 : SUCCESS in 1h 27m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87423d9dedff49f585ffbe98b15cc01c : SUCCESS in 1h 00m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106158c969444bc58b9a893c4ed5a821 : SUCCESS in 14m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e20aa5809dc74b0381e3df27b37f2c74 : SUCCESS in 22m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf7f69a220de47a8a354ddae36bc7c20 : SUCCESS in 2m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40010fe31a06416dbe4300a4691ab8b2 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/909bdb8d6b33485c8a639ef56e966e21 : SUCCESS in 5m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27836133d7724f99941a74410a6489ec : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d09c98db7c44e18bf39cdba09d20e23 : FAILURE in 38m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/753c41dec4ae42198d94785ebba9199d : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3507e6cd48c476bb2398e589dc1672a : SUCCESS in 1h 44m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b1ae99786aae42dfa78057a17b89a6f4 : FAILURE in 2h 45m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07f9b76db17a4fe6a24498e1ec03319d : SUCCESS in 2h 02m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/161f7dc82bd240098755a35cc4718418 : SUCCESS in 1h 51m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1ad1221be0ce465281188d5c30800bd7 : SUCCESS in 39m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/675fa43b20424afd8a664e42c08cb633 : SUCCESS in 1h 12m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9a2c25ca8c641dda32cea33d28dfcb6 : SUCCESS in 1h 47m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f2c1877a7df243e4a2cd645dddf3d2b8 : SUCCESS in 46m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d4daa1bc3d34a32b2b4576a6d58d441 : SUCCESS in 1h 54m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6db19fe916de47da91443a46847d2cb3 : POST_FAILURE in 2h 35m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1902552801a4f5aab38341a0d14e703 : SUCCESS in 1h 39m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44b0583cb93c4fe2ba35b7206043be51 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":16},{"id":"0efe528f46786ee3b298bc790b7b74fef0a68e2f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-11-28 09:56:02.000000000","message":"Patch Set 16:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/16/20251128084826/NEC-ISCSI : FAILURE in 10h 02m 27s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/16/20251128084826/NEC-FC : FAILURE in 10h 07m 32s","accounts_in_message":[],"_revision_number":16},{"id":"c10ea6d1f38102b84737fabef5d7f0147f965b4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-28 12:45:20.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"9dea5de31835f21176a8507e82085a92ef677d09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-28 12:45:31.000000000","message":"Patch Set 17:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/81/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"1e20a2123cef7cae61bd8dbbd479382759afd017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:10:07.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/975cf04496e9497a9a139141db1736eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df2e0a296b954359964e53bc0a998c87 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5bb0ce9ce2a4d0582fd406121619bc5 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a4dfb9f7ede42ddbfd0240d954cfe26 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0ce3d943d404081dc15576daeba7ecc580c02450","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 13:36:44.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2585/.","accounts_in_message":[],"_revision_number":17},{"id":"b49c5ee1fb696ac8e6c514cf9b0daf3a2fe758aa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 13:41:36.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2733/.","accounts_in_message":[],"_revision_number":17},{"id":"5187b3a1acbdc4ac5cac85b42f720831f5d59528","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-28 13:44:48.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2bba41cef8204ceb8d1fb2ed9c1a05f2 : SUCCESS in 55m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/70254315be994d16a9beaeaa16c32d4c : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":17},{"id":"7b9e44f1bb7bd5e6d784702f02a6019a919b4e36","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-28 13:52:21.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":17},{"id":"891485558c126b9a039a9785f23fe0fabc774c5a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-28 13:53:11.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 05m 48s","accounts_in_message":[],"_revision_number":17},{"id":"eecc32d09bd18c9f5498a808fae17d3b7c052ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:16:54.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e81f90fb6804ad3b5727bf3d9e3f4e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89010e0589d14929bf9f4020048770ac : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e2590cc09b94b3ab7e9d515f5112a8d : TIMED_OUT in 40m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/568a541a6a1046a38ca71189140702c2 : SUCCESS in 14m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6aee4568c134d94831c222e257146db : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/58ab0525aa3a44c9bd5e3cfbbee1e3b0 : SUCCESS in 1h 05m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/254ba8ea2cf547d3aab3affdf6fd2a6f : SUCCESS in 28m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca474314beb64ccc9c0b35f89795d14e : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a58f3643e5643e38b376c6849400ce0 : SUCCESS in 32m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc2d7db752c4290b0622ddb12ab3b70 : SUCCESS in 12m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2c65f304c9604e7596fe8b786a199312 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98abf003cf0c4b628dbf5809201733a4 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/74f50ad2f1dd49058ca650db7e76c531 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16cc2e5d4b7e49b8bbc7df3589afca1d : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f5ef4b4fc56c44aa95ee058e9b38e2fe : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/509d471ad3e447a185318c86ad3dac4d : FAILURE in 22m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37dbe2a770654777abe66e9af7affe0c : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88833bd472fb4086a48ec4e30d07dda5 : FAILURE in 1h 34m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ba97606514cd4f82b1c9d136ad82b752 : FAILURE in 2h 23m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/330a8576c128498fac3aa983cda449a7 : SUCCESS in 2h 21m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/de5923190353448198e817900b31fbd4 : SUCCESS in 1h 22m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/67dcc539d11e47a2aa2f33ba80a9c595 : SUCCESS in 20m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a166ab6ecd84c32a7726fe98fdab704 : SUCCESS in 1h 26m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/573e072bc7c44ba1929ed6e99c983839 : SUCCESS in 1h 41m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd2025b7cd2247a58f38961774c8a42f : SUCCESS in 35m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/80890f8134684f77842ae8b1b50753eb : SUCCESS in 1h 10m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/efc8bcdb19cd4ffd9414493dc9357502 : FAILURE in 2h 01m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5a50d1517414baea7dc456e041adc4b : SUCCESS in 1h 16m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6f949a10c85413e946d8c9bb8556159 : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":17},{"id":"bdbe2ca6f5ccb65bd92d740bb01687c63a2af8d5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-11-29 06:20:08.000000000","message":"Patch Set 17:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/17/20251129051001/NEC-ISCSI : FAILURE in 9h 47m 09s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/17/20251129051001/NEC-FC : FAILURE in 10h 10m 04s","accounts_in_message":[],"_revision_number":17},{"id":"e47a812434e7ba217f95ad310e547ff4a69d992b","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-01 09:13:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"230ef53bbe4c4c479ff88a16fe89161066a2b9f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 09:37:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7ea40e0bf7411b9cb6d178ab4242b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cfd497b9c750484181c6931475edcbac : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e2fd1377dce4573adf0859b4ee774d5 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f60a33546a544789c3cd5544d9aef27 : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"078f4fc4498870c96c4770f508a4f57087fbbe03","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-01 10:10:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bb46de351109471b8df1e1c39be10840 : SUCCESS in 49m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/483a5f772103410da0f064d227b9ab91 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":17},{"id":"2c3f9b61689d33b9dcca4cee597e24deaa08a2ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:00:30.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad2678b15cd4cf4b59b4659ead7c16f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864a01b7337d45ada365acda14e61cbd : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2513b01211b6486c97c98a505ebfbc9b : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e0e184cd51d45ac9d56fbfa27275136 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25f78c8c3fc482485992e786de6dec7 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/f2c162f1ad22433ba0fe8c1a5d16d056 : SUCCESS in 32m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7efd69795f9a4525800021c68b036ac2 : SUCCESS in 59m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e89a1b7600d0468b9e29c517e510dd7d : SUCCESS in 1h 34m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac41cbddaef844b886cc9833de4aad6d : SUCCESS in 1h 07m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25aec7d735ea45cca5fffaeab1b88af3 : SUCCESS in 18m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca01b7e6360c4c53ae15270cdb8626b4 : FAILURE in 26m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6584684e616244da9b9d6069455a51f8 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/437ede950c844759aa7fcbcdee1a1bdf : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5fec8208a6f64e8692fe41b638329fa7 : SUCCESS in 5m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8ef3de49d1da413680b4f6c98b01f2dd : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9798658e8a743db9472429b7e6da501 : FAILURE in 38m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49262e3a14ce4938b2233f5bbf9d3fc4 : SUCCESS in 4m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a5c4df693eae486faa387b7308bc682b : SUCCESS in 2h 01m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2fdb9c321a414f7f8b226ca736cb9968 : SUCCESS in 2h 37m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/882d23ce6e944241bd9088c46d0ead1e : SUCCESS in 2h 15m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/767304b0c4464d18b0fb1c971d8c544b : SUCCESS in 1h 54m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b0bc1e1fbb434429bff11d67ee3f88fe : SUCCESS in 36m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e63d45a412d4be98c436f2bd2fd03dc : SUCCESS in 39m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4548c02532964fdc828eff7802dc33bc : SUCCESS in 1h 11m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ec28fcec391f46d1ab44205499195af0 : SUCCESS in 46m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f7f8b5f9c2c4efb987718857b575445 : SUCCESS in 1h 36m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ca99ba2cc23047e19e8914a1fdbbdbe7 : SUCCESS in 1h 46m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d7e675fb925749df8293ef33471865ff : SUCCESS in 1h 35m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4f23b3a2a245559c729c4f6f2cc95e : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":17},{"id":"d3485b0c689e01e1d7af64792fb1dbdaea8e0841","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-05 10:47:06.000000000","message":"Patch Set 17:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/17/20251205151504/NEC-V-ISCSI : FAILURE in 3h 29m 56s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/17/20251205151504/NEC-V-FC : FAILURE in 4h 31m 59s","accounts_in_message":[],"_revision_number":17},{"id":"55d82e0e678f0c602b68c1ac00b42a53e7fbbe0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-05 14:04:40.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"032048b606b4781099c2cc3e9c74bfda9c26c9ed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-05 14:05:02.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2636/.","accounts_in_message":[],"_revision_number":18},{"id":"b0028a8b0aa6add80ac4a26a898eff5bb03fac2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-05 14:05:02.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2784/.","accounts_in_message":[],"_revision_number":18},{"id":"68b107a676bd9c376ad8e568dbfa054c92da1b19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-05 14:22:07.000000000","message":"Patch Set 18:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/135/ : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"e0c2ff72baff1a5aa4f56e6083b71b6717cead99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:37:47.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d7395fcb384dc980f9ff1aebc37f0f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25449014524f4cf7ab63d2e9c9eacad1 : FAILURE in 17m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/688d052c6f464003a587370b0e7ae28b : FAILURE in 12m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/921e18ae22ba40308a12d640512af3c8 : FAILURE in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5c4e82e1458437b9e013be7c85bd30ff332630eb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-05 14:59:55.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":18},{"id":"6a608a9cd1fc30a37dd8d3ee2a50b63a2f7d2eb6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-05 15:24:35.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/50882e671b9f4be8ba87c5a5edf5f8d9 : SUCCESS in 50m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/caaeff65ab8340d88c4755cc3693f91f : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":18},{"id":"21e634a4b0866dfa12554580532c8638594c5335","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-05 15:48:20.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 09m 25s","accounts_in_message":[],"_revision_number":18},{"id":"21e02af6e6d5a2b11146b894fe3cb27170725d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:27:52.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ec1d02c628d4d71925e83bafab8558d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68ef3d8083f406ea447325f2396059d : FAILURE in 7m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b805edcd8d643708b0fd13707224445 : FAILURE in 12m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d797e2b95564c1aa900a1536588b10d : FAILURE in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1abb19691a4f058adab8b4c40d9b51 : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c3ed249d22c74afd9add8dba56cb2945 : SUCCESS in 47m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6abd7d29b5204262ac661c892233e53e : SUCCESS in 44m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d180364d63c424fb14d0db502774d12 : SUCCESS in 1h 24m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d797d5b5829c427e998b640737c17240 : SUCCESS in 1h 00m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0cfc9f6a6834028b8a24de9fc2d9b1e : SUCCESS in 15m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3989855a7b4b43bcbc6279eb0cec8732 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3b393cf03f74ce08e65f46e75cf34aa : FAILURE in 6m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bed0209fc87b4abfbe8665c9b9152a67 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b4d97d78e6a45879d41428d0899e0ed : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/149d746d4f3b45a2a8768b2b145dd2fa : SUCCESS in 8m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3f75d5d696a4983b83b78f837a2fbd6 : FAILURE in 40m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/591d2bf50076442dbc3768e0a77a10f9 : SUCCESS in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5f93170062042218f8cc7b79001b0ae : SUCCESS in 1h 54m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4b7cb3ec076a44248b355c3b7c917ba3 : FAILURE in 1h 33m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5e02ab366e94ae5a6ad41d70782edb4 : SUCCESS in 1h 55m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ca8092eef75d44ad855c509f7577fc6d : SUCCESS in 1h 56m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a51025f3e5474929b774e435c12457f6 : SUCCESS in 36m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/946dfe39216143a6b2ebb2f9d21494a4 : SUCCESS in 1h 18m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3569be6028474e9ab7a73d2829030301 : SUCCESS in 1h 33m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dadbb7215c2d41b6a8f0cb1e15e49b1e : SUCCESS in 43m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/339bad3de69247b880922673c028a339 : FAILURE in 1h 44m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0d57b02a76f04855a433237c6e7f6ebc : SUCCESS in 2h 17m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f3052f377144a12b67ac23f9f035be4 : SUCCESS in 1h 40m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8524b5bce2c4a1bab4c47ccf5464964 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":18},{"id":"a19467a501ab00a01287f77f7d402e2ff7adda30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-11 10:06:22.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"c724fd95e203bd9b390118c45d71f12907a69018","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 10:16:51.000000000","message":"Patch Set 19:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"89f1a8551f193b4f795de6c91694b3a1ca5de44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 10:16:52.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"0d891b1e8e72112ddc0fe24eca9b573537d68e35","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-11 10:17:09.000000000","message":"Patch Set 19:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"efe09242bb5b29405c4adeb5f5e600a4555f3f44","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-11 10:57:50.000000000","message":"Patch Set 19:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2683/.","accounts_in_message":[],"_revision_number":19},{"id":"f57bcf375be144089397ddf5f6e00c3f58202e95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-11 10:59:05.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2830/.","accounts_in_message":[],"_revision_number":19},{"id":"d4be1cce9de1a4d4de794e6ecf9d2e627494611d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-11 10:59:12.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/572e40a059af44e880250ab35717614e : SUCCESS in 49m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dd0b548bc57c463cbc433906288485fe : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":19},{"id":"51094fb9867d9e7a104d49e77d09ba72c803a872","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-11 11:09:10.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":19},{"id":"b46b8e972622f808634ea2ed0ab694950e81d994","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-13 08:55:24.000000000","message":"Patch Set 18:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/18/20251213133343/NEC-V-ISCSI : FAILURE in 3h 34m 08s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/18/20251213133343/NEC-V-FC : FAILURE in 4h 21m 38s","accounts_in_message":[],"_revision_number":18},{"id":"dc632680306ccdf4d2c4350ea64f788b149e9780","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-13 23:50:35.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-18-fc : SUCCESS in 58m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-18-iscsi : SUCCESS in 1h 6m 48s","accounts_in_message":[],"_revision_number":18},{"id":"9f47b542bd1645dc5f013d95625d4f04adc82a81","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-12-15 01:10:07.000000000","message":"Patch Set 18:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/18/20251214235250/NEC-ISCSI : FAILURE in 9h 39m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/18/20251214235250/NEC-FC : FAILURE in 10h 17m 14s","accounts_in_message":[],"_revision_number":18},{"id":"06eb1f2cb85d6177ba4cf80a5755f07785428fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 10:33:20.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"662621bf9b5f9adc91129fa3f823f4a41d37f0fe","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 10:37:32.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"097dd8e14b3e0ded1eaa47f8cb2caaa1c067db8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 11:03:49.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6406cc24ce431a8858518a4814a465\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89b52c4cbb0d4affa5e67b1f531791ec : FAILURE in 12m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cfcedd87f0d4428ba968b84c19fc662 : FAILURE in 13m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d92f97f405624bf48d622d890b7523c8 : FAILURE in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"21050c77cc0b3f18d146c2f1300a7012cce8afb0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 11:30:00.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2891/.","accounts_in_message":[],"_revision_number":20},{"id":"df7a8a24d0c76a9e109accb68009868108f33219","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 11:30:02.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2744/.","accounts_in_message":[],"_revision_number":20},{"id":"55414d2e125700bc3061a71a8b04fee31fc8abe9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-16 11:34:41.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":20},{"id":"2827c200f1734c827f755f063f409cbdfdb72e37","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-16 12:39:56.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d3a772841d794146950f3fa66b61d7d6 : SUCCESS in 45m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1496123e31a44953be4c0973a9baac8b : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":20},{"id":"c2fef7124cc19bb5eedbfa615e52a4b0c59a1dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 13:48:21.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4d3e31aae954edfb7cdd981b428639b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffe436f07b14c2e8d0b4e059b7f5664 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7df35870d4d54be89629cf26a8e777f4 : FAILURE in 9m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/feb6f7e090b24d6fafbeacb793f55cdd : FAILURE in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a9cc67fda04bcfa4454e0657c46207 : SUCCESS in 16m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/679d82fe270343a3ba718a66b45f82dc : SUCCESS in 35m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ace8155ebf7643a0bf4e6775852a6a48 : SUCCESS in 28m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd6366f9b889456badecc9ddd9cc37f8 : SUCCESS in 1h 44m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b54d1dca0bf45519142557c7cc88952 : SUCCESS in 1h 01m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2707defe6514e458d443db869c6de76 : SUCCESS in 9m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f838680ede34c56b04692d8cdc06454 : FAILURE in 21m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e9fbdb771db4d8ca9aa106a805f98b8 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73f7ddc41bc84b86b301dd102323136f : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16bac532e664485088cb0f3d9ab84e65 : SUCCESS in 4m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b5dedc4338846a09d339461b609b135 : SUCCESS in 8m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/16030896ca4049218720ce682d337450 : FAILURE in 20m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eec3d42d92ef4e208f680d9c35046dde : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b544f15587b45678e66017cde86f981 : SUCCESS in 1h 47m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e03cebb424e4b32bb9a30d2975dc237 : FAILURE in 2h 03m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95122f7cd210482a9396576637a990ec : SUCCESS in 1h 36m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ce2abf3ead1d4c69bbf6b33520c0a291 : SUCCESS in 2h 05m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/df791dd0d0224edfb5b6dd48ab78994b : SUCCESS in 33m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e64023653274c0f95604f13f640d615 : SUCCESS in 1h 09m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fc3f20f41605420ca7fecea304fff232 : SUCCESS in 1h 48m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2bbaf6b5ee0d46e2bb390e9173b8a34c : SUCCESS in 43m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aac608b17ccb4e28bf33266fbd3752d0 : SUCCESS in 1h 53m 03s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e3a2fa5cc95647739509d0c35cd2720b : TIMED_OUT in 3h 09m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/addf7a26e4e54875983d601d1f6ef65a : SUCCESS in 1h 35m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fa2aa69e3fe44f3889be7b6320fc23b : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":20},{"id":"6b18d79b431ff465c5cfec5354a46d336ef83698","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-16 13:55:23.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":20},{"id":"fb7f6e33ac00b391b95c24fd816b2bbb5b5896f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 14:25:07.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"b408eb4e14f6ffeee7e6ba80ac8c29c6df77a71f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:47:35.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f612388afc084e278f810d5d49ed6e8f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb76d3edc68645e59988c58847b354de : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d120069129bf43f6861614519aabf27b : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44af1cc39c9741839731ecd63395f0e2 : SUCCESS in 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a14676e59379f207162ad8012426be671bb0f9e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 14:58:37.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2751/.","accounts_in_message":[],"_revision_number":21},{"id":"314329ab8cfe79b25c85e16efa7ff31b19e030b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 14:58:42.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2898/.","accounts_in_message":[],"_revision_number":21},{"id":"155585b83019da8d1a1837e31a764fb7c211cd9a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-16 15:22:00.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":21},{"id":"2b3c7b9b9e911a24b952b8091e59e9dc3fce89ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:37:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c405d91a52421eb16eeb91d9d4c438\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc8a25a0447469ebebf314d410d3d1f : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6688e31c2684a1893b56af67e6f879d : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd254dd98bd49ddb0023650f5a92777 : SUCCESS in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81c9a54fa21480596952f90c301fe21 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/0beb253f2f9b4d1d89f23a459338b739 : SUCCESS in 1h 12m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47879edb3d5840839853f841b5437483 : SUCCESS in 57m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b2a3702e6564fc48e9add3d09b2d439 : SUCCESS in 1h 34m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc481d9f24df470fa61a5909105f3332 : SUCCESS in 1h 01m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/596ef7e45c764ff0ade583f9e2565202 : SUCCESS in 11m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8194b21f7862484980de2e0eff737b02 : SUCCESS in 19m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0b1335507f7c4b92a9d53dbc7426a9a6 : SUCCESS in 6m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4c374690294444986eaffd3ec22b87f : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/202a1a7d19394015b069f653930c395a : SUCCESS in 5m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a4a8f8aed4874ebe8d6cdfef2c35504e : SUCCESS in 8m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2fc187f98b2541b186ef62be21d7c7e7 : FAILURE in 29m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8a89969773340859cc633b2615975e2 : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f844e39d47534a0d8212a519da1fab24 : SUCCESS in 2h 07m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/adc31a16da8a4dfea433777c0f316bf5 : SUCCESS in 1h 22m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d4404dc6d3643fd904fe047a9673e40 : SUCCESS in 1h 50m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c1fea17b05734acfb310bcf890704167 : SUCCESS in 1h 30m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/62f9c8ea7527428cbde21468e94136d4 : SUCCESS in 35m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/62edf9982d734bb59ffcec274c81e687 : SUCCESS in 1h 00m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31c80ac18a564206aacf47e6e91b79e4 : SUCCESS in 1h 29m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b56be2f08dc49d79deda18b0b0edcac : SUCCESS in 24m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6f203f2128d44c4b9e4f8ce2845610c : SUCCESS in 2h 09m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d3f87621662f41389c9dca4958c2cda2 : SUCCESS in 2h 07m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e8bac800e834271afa9a2a73e85d6a1 : SUCCESS in 1h 46m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a32227647a49b99f5b40282a112ca1 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":21},{"id":"ce0665dd91c1a7eeb33fda7a653a0773f948c75b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-16 16:44:25.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_12_2025_16_37_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":21},{"id":"9a27e6da060eb70e4337ab4ef88779c128564dd1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-16 16:47:53.000000000","message":"Patch Set 21:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"d1f4d14a9a8ba15124a443791ae82b25b5c9642d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-16 17:16:12.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9db79a56418a4a588bee07748e6ddc2e : SUCCESS in 48m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7833fc9e8f324ad68128127820d3f0e3 : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":21},{"id":"89e569edc6f360f562987aab20ecbcaa59e773f1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-16 21:15:05.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-21-fc : FAILURE in 1h 19m 8s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-21-iscsi : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":21},{"id":"93369d6de6115de5858a8d5996cee281b8b5f204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-20 15:06:12.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"d7b72f82b7ace23801f2ddeaadeb44160131ce90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:37:59.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00dd89dc9d6c4fb28ddb1f700d70d159\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d90aea46c3774dfbbf45b61f8abe86c2 : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ce3f56e57ee4d7a8b50613b8ac1e6bf : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a59c1bd996f4caeb7b68d2fde1e2326 : RETRY_LIMIT in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"1d53bc4f76636c97c7dc1cf211417135aa2c6e41","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-20 15:40:19.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e9eb2281367f40fd9e5b187f2fe8a3be : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":22},{"id":"222915060071d6292a31274e340f48f37cbcb85c","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-20 15:45:17.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/29fda670c0574cd69ba1e0e746c6f964 : FAILURE in 38m 10s","accounts_in_message":[],"_revision_number":22},{"id":"fa4abfcf49ccec91ea71d7ac5bf274a347058c67","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-20 16:08:51.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":22},{"id":"3fcdeb2ac9c84fcdb582794a90c9a06a7f92ca47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:25:41.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736e1a09b9534a72ab3b3595ef1dfc68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e697b17eb144489af176609a0b0f04c : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb58d36170274674a93ac96063b5ad02 : SUCCESS in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95929364d82d4d0fb92582cc83760dbe : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84095b7d76f74e33931c1d8648b804c5 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c58ed030f9a44b93b09fdb63e71d11ec : SUCCESS in 48m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/052c71a20afb4f24b402c2eb49b70f27 : SUCCESS in 1h 04m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cdccad6238a4005a102c7a3f85298f2 : SUCCESS in 1h 35m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0945253aa6d14f24871055ba2f0a19fb : SUCCESS in 40m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec61ca4c5dbd44f3b85e4ace196ace6a : SUCCESS in 13m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e3e9841693d4ae2992e41add450ce3f : SUCCESS in 24m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f2674f04ff04d12a2edbb76440205d8 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3209ee5a2e08409f872a0780f21e1664 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae585e04df6b4fefb13044fdc1e4f96c : SUCCESS in 6m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5dcc9310edb1434b9879078e373f5454 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea9ff9b1b3a94288864340e951529dcf : FAILURE in 43m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/148b3b42e12f4c4aa868214687c6c60a : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/781f538780a644098ee524667e5ba836 : SUCCESS in 1h 43m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6160fae145584f8fbe98b63913bb0add : TIMED_OUT in 3h 05m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69d2131c9b4746d8bc997ad14a3208fc : SUCCESS in 1h 50m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8cebb61dc18f4ca984c973e715cc1183 : SUCCESS in 2h 14m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e6c0bee14a034c9da62ab7b5490be306 : SUCCESS in 35m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fcaaa63551a14f29b88bbe1a028ffd51 : SUCCESS in 1h 03m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e4ab814fafbe4e01b1d85a3c1800e3a9 : SUCCESS in 1h 27m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9cc511da650248d686b5ed3ef7f882af : SUCCESS in 31m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c919e0641e1442781f86c8272f73f28 : SUCCESS in 2h 06m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5b046e4b83a74a88b6fe87b1fc16c4b6 : SUCCESS in 2h 18m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00ce50033b1e4ec0b16e85db68cf285d : SUCCESS in 1h 44m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/722faefae778423baf36bab6f8ad9087 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":22},{"id":"30f5dcf61d6dfb6fcdeaa2d4609931871371b266","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-20 18:32:31.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_01_2026_18_25_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":22},{"id":"938f5a920ee6fcbcaf7bca6522eb35b611d56bf0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-20 21:09:39.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926298 : FAILURE in 1h 18m 54s","accounts_in_message":[],"_revision_number":22},{"id":"6751a2426513e643a2f908fcbdda1f14d1436c48","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2026-01-21 01:19:16.000000000","message":"Patch Set 22:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/22/20260121060545/NEC-V-ISCSI : FAILURE in 3h 22m 26s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/22/20260121060545/NEC-V-FC : FAILURE in 4h 13m 28s","accounts_in_message":[],"_revision_number":22},{"id":"00eb6a622f6bf9d8d5c4622ee9bc5a8c98a7f559","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-21 01:57:58.000000000","message":"Patch Set 22:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_308/index.html : FAILURE in 131m","accounts_in_message":[],"_revision_number":22},{"id":"223c2e00a7fd73b6c484a80cb138f88194a165f4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-21 03:40:29.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4511c72e483440de940dcd32939e8290 : FAILURE in 1h 01m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a592c58458944bf1b18dbd0f0edfb5b1 : RETRY_LIMIT in 14m 55s","accounts_in_message":[],"_revision_number":22},{"id":"4b2c7fdd37f299ec04f6b42636d8d839f6edd8ff","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-01-22 01:40:22.000000000","message":"Patch Set 22:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/22/20260122002306/NEC-ISCSI : FAILURE in 9h 39m 26s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/98/926298/22/20260122002306/NEC-FC : FAILURE in 10h 17m 12s","accounts_in_message":[],"_revision_number":22},{"id":"067295204d43ecca653a51d7e80ad643024ad920","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-26 21:30:29.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-22-fc : FAILURE in 19m 37s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-98-926298-22-iscsi : FAILURE in 2h 14m 1s","accounts_in_message":[],"_revision_number":22},{"id":"b220c2fa15f2e4056243043485f8315f4cc8c464","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-04-20 07:54:02.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4716ef4a922e47bdf4b57b3623bc2d4e65f21cb8","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-04-22 08:26:12.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"24cff6f789ee5e8bb4c10e5b269f36ea3f802812","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-04-23 08:57:03.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"675532bff26eaa9667dbc4761e4abeb8b96e24bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-06-10 14:49:10.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"492351b1fefbcab42e2a7dcbbf6c04afc1523a9b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-10 15:12:05.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"95e2a2832641f8434fb2902fe71c1521c633ac52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:12:06.000000000","message":"Patch Set 23: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"5032742e7f8cedf6203536438810ef012209f889","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:12:15.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"a85680fd5dbac95cfc7af6bc0e02340e39439634","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-06-10 15:12:19.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":23},{"id":"d957c9795be5adf8762f302a6ac1ad2ad6d7dd29","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-10 15:12:21.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"d549519911a459238ee607604e89f4d3568281e2","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-10 15:12:24.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"c6b359545fee83d3e428d8e41cc82875da269c11","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-10 15:12:24.000000000","message":"Patch Set 23:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/926295 is needed.","accounts_in_message":[],"_revision_number":23},{"id":"0d4dff0c320092e13852a4ba1a11711154bec360","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-10 15:46:27.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefa3ecc8e68a832f88d7f63ba2734299587d908d .","accounts_in_message":[],"_revision_number":23},{"id":"ed964ac62c8323a3b4a96051ccb1d5f7955c34e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-10 16:46:22.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/8f5aa8c9d89047d7bcef8254326cfc40 : SUCCESS in 53m 49s\n- cinder-storpool-tempest-iscsi-multipath build/86144bba0b7d40d1ba400bb5ae39c45e : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":23},{"id":"b4d1e2b7dc921957c24f86f373efac05a9b9f1e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-10 17:26:47.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4421/.","accounts_in_message":[],"_revision_number":23},{"id":"175ff8ad255cdd9766ad879aefe6ecc7617d3c36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 04:30:47.000000000","message":"Patch Set 23:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4245/.","accounts_in_message":[],"_revision_number":23},{"id":"40729a82c4ee4626bda2c5f817774e640fd606d8","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-09-02 09:15:09.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"6077a82b72b1f7607c57c5b915773cadad973357","revisions":{"7f0ee6c5d1df539be012e6c8374773c8b7d42e6d":{"kind":"REWORK","_number":1,"created":"2024-08-14 15:59:14.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/1"}}},"commit":{"parents":[{"commit":"57db6811efd014f3c2af85e5a67865876cd0c851","subject":"Merge \"PowerStore Driver - Add a unit test for connection properties for NVMeOF connector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57db6811efd014f3c2af85e5a67865876cd0c851"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:59:11.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nReplaces \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 by \u0027os_encrypt_key_id\u0027 and\n\u0027os_encrypt_key_deletion_policy\u0027 respectively. Adds the\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 attributes to encrypted\nimages created by Cinder.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f0ee6c5d1df539be012e6c8374773c8b7d42e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f0ee6c5d1df539be012e6c8374773c8b7d42e6d"}]},"branch":"refs/heads/master"},"757b1ea7c29b047fda29f92b356579286242716c":{"kind":"REWORK","_number":2,"created":"2024-08-20 15:06:35.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/2"}}},"commit":{"parents":[{"commit":"57db6811efd014f3c2af85e5a67865876cd0c851","subject":"Merge \"PowerStore Driver - Add a unit test for connection properties for NVMeOF connector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57db6811efd014f3c2af85e5a67865876cd0c851"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-20 14:59:47.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 for encrypted images\ncreated or imported by Cinder. The deprecated attributes will still\nbe processed if present but only the new ones are now assigned to\nencrypted images created by Cinder.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/757b1ea7c29b047fda29f92b356579286242716c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/757b1ea7c29b047fda29f92b356579286242716c"}]},"branch":"refs/heads/master"},"98c12dda7196dee37c878cabb364377674c87a1e":{"kind":"REWORK","_number":3,"created":"2024-08-27 15:30:17.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/3"}}},"commit":{"parents":[{"commit":"57db6811efd014f3c2af85e5a67865876cd0c851","subject":"Merge \"PowerStore Driver - Add a unit test for connection properties for NVMeOF connector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57db6811efd014f3c2af85e5a67865876cd0c851"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-27 15:29:46.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 for encrypted images\ncreated or imported by Cinder. The deprecated attributes will still\nbe processed if present but only the new ones are now assigned to\nencrypted images created by Cinder.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98c12dda7196dee37c878cabb364377674c87a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98c12dda7196dee37c878cabb364377674c87a1e"}]},"branch":"refs/heads/master"},"d6ca97bc79dc73a0a17b1f8b5eb8100c77bc9a7c":{"kind":"REWORK","_number":4,"created":"2024-08-28 08:47:24.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/4"}}},"commit":{"parents":[{"commit":"57db6811efd014f3c2af85e5a67865876cd0c851","subject":"Merge \"PowerStore Driver - Add a unit test for connection properties for NVMeOF connector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57db6811efd014f3c2af85e5a67865876cd0c851"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-28 08:47:13.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 for encrypted images\ncreated or imported by Cinder. The deprecated attributes will still\nbe processed if present but only the new ones are now assigned to\nencrypted images created by Cinder.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6ca97bc79dc73a0a17b1f8b5eb8100c77bc9a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6ca97bc79dc73a0a17b1f8b5eb8100c77bc9a7c"}]},"branch":"refs/heads/master"},"9bcf919c30518284632dab42ac4a13c19bcda338":{"kind":"REWORK","_number":5,"created":"2024-10-28 16:27:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/5"}}},"commit":{"parents":[{"commit":"57db6811efd014f3c2af85e5a67865876cd0c851","subject":"Merge \"PowerStore Driver - Add a unit test for connection properties for NVMeOF connector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57db6811efd014f3c2af85e5a67865876cd0c851"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-10-28 16:25:48.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9bcf919c30518284632dab42ac4a13c19bcda338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9bcf919c30518284632dab42ac4a13c19bcda338"}]},"branch":"refs/heads/master"},"a376e663ff2d232d3e11bd9c7f23860ba3016eab":{"kind":"REWORK","_number":6,"created":"2024-11-12 12:49:17.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/6"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-12 12:49:03.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a376e663ff2d232d3e11bd9c7f23860ba3016eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a376e663ff2d232d3e11bd9c7f23860ba3016eab"}]},"branch":"refs/heads/master"},"e1c844fd50a88cf58d48b40179eb601cfd0aae9d":{"kind":"REWORK","_number":7,"created":"2024-11-12 13:23:51.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/7"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-12 13:23:38.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1c844fd50a88cf58d48b40179eb601cfd0aae9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1c844fd50a88cf58d48b40179eb601cfd0aae9d"}]},"branch":"refs/heads/master"},"542b808acd9d1772ce0142ce87aaa7252bc95726":{"kind":"REWORK","_number":8,"created":"2024-11-12 16:29:40.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/8"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-12 16:29:08.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/542b808acd9d1772ce0142ce87aaa7252bc95726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/542b808acd9d1772ce0142ce87aaa7252bc95726"}]},"branch":"refs/heads/master"},"6b352d908f050a5ae5acea47761e90353ec6c274":{"kind":"REWORK","_number":9,"created":"2024-11-19 17:12:09.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/9"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-19 17:11:45.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b352d908f050a5ae5acea47761e90353ec6c274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b352d908f050a5ae5acea47761e90353ec6c274"}]},"branch":"refs/heads/master"},"2cfb2a8bbd9a3c054fd49ac85ee40939fdf3add9":{"kind":"REWORK","_number":10,"created":"2025-02-17 13:58:56.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/10"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-02-17 13:58:36.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cfb2a8bbd9a3c054fd49ac85ee40939fdf3add9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cfb2a8bbd9a3c054fd49ac85ee40939fdf3add9"}]},"branch":"refs/heads/master"},"901913bb29f42e9a8fce4d4fbbfb5463f4683a2e":{"kind":"REWORK","_number":11,"created":"2025-02-24 09:54:33.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/11"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-02-24 09:54:17.000000000","tz":0},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/901913bb29f42e9a8fce4d4fbbfb5463f4683a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/901913bb29f42e9a8fce4d4fbbfb5463f4683a2e"}]},"branch":"refs/heads/master"},"36b72dd35bf1923a5428e79db1aedb91f3118c11":{"kind":"REWORK","_number":12,"created":"2025-02-25 17:50:44.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/98/926298/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/12"}}},"commit":{"parents":[{"commit":"6e2884468d8299c0eee520bfcfffd24431585ec0","subject":"Merge \"Fix \"signature_verified\" metadata propagation to images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e2884468d8299c0eee520bfcfffd24431585ec0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-02-25 17:49:37.000000000","tz":0},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36b72dd35bf1923a5428e79db1aedb91f3118c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36b72dd35bf1923a5428e79db1aedb91f3118c11"}]},"branch":"refs/heads/master"},"b1a88429125ebac5ec4776444dff1f8b76c19fab":{"kind":"REWORK","_number":13,"created":"2025-05-21 14:36:55.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/13"}}},"commit":{"parents":[{"commit":"d23a17da679e1180a546d78138e257f397fe4b4b","subject":"Merge \"nfs: minimal mypy coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23a17da679e1180a546d78138e257f397fe4b4b"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-05-21 14:36:33.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1a88429125ebac5ec4776444dff1f8b76c19fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1a88429125ebac5ec4776444dff1f8b76c19fab"}]},"branch":"refs/heads/master"},"010fcac43c36b1925087c64a94c80163949df4f7":{"kind":"REWORK","_number":14,"created":"2025-05-26 14:33:30.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/14"}}},"commit":{"parents":[{"commit":"d23a17da679e1180a546d78138e257f397fe4b4b","subject":"Merge \"nfs: minimal mypy coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23a17da679e1180a546d78138e257f397fe4b4b"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-05-26 14:33:14.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new disk_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/010fcac43c36b1925087c64a94c80163949df4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/010fcac43c36b1925087c64a94c80163949df4f7"}]},"branch":"refs/heads/master"},"546ee01cc2d0e471e85601f1186fa60dd64137cf":{"kind":"REWORK","_number":15,"created":"2025-11-18 14:51:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/15"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-18 14:49:25.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/546ee01cc2d0e471e85601f1186fa60dd64137cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/546ee01cc2d0e471e85601f1186fa60dd64137cf"}]},"branch":"refs/heads/master"},"7093a87778a0c9df9c7ae022e72c9bf5a1ab178e":{"kind":"REWORK","_number":16,"created":"2025-11-27 15:26:01.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/16"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-27 15:25:42.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7093a87778a0c9df9c7ae022e72c9bf5a1ab178e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7093a87778a0c9df9c7ae022e72c9bf5a1ab178e"}]},"branch":"refs/heads/master"},"66bbaf67588bf3723465ba6bc5cefeda58c3edb7":{"kind":"REWORK","_number":17,"created":"2025-11-28 12:45:20.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/17"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-28 12:44:41.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66bbaf67588bf3723465ba6bc5cefeda58c3edb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66bbaf67588bf3723465ba6bc5cefeda58c3edb7"}]},"branch":"refs/heads/master"},"7e6cfd52ee98389da542f5f6fed2f4d3e62b96a8":{"kind":"REWORK","_number":18,"created":"2025-12-05 14:04:40.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/18"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-05 14:04:21.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e6cfd52ee98389da542f5f6fed2f4d3e62b96a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e6cfd52ee98389da542f5f6fed2f4d3e62b96a8"}]},"branch":"refs/heads/master"},"ce0b1e527d61cde8892af46f7e02f1c344e8d76e":{"kind":"REWORK","_number":19,"created":"2025-12-11 10:06:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/19"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-11 10:06:09.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce0b1e527d61cde8892af46f7e02f1c344e8d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce0b1e527d61cde8892af46f7e02f1c344e8d76e"}]},"branch":"refs/heads/master"},"c7b9d5ac7244ad894fa95df62d6863466e709043":{"kind":"REWORK","_number":20,"created":"2025-12-16 10:33:20.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/20"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 10:33:09.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7b9d5ac7244ad894fa95df62d6863466e709043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7b9d5ac7244ad894fa95df62d6863466e709043"}]},"branch":"refs/heads/master"},"77b3596f79af0cdb442286e6d24642407ed1600d":{"kind":"REWORK","_number":21,"created":"2025-12-16 14:25:07.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/21"}}},"commit":{"parents":[{"commit":"77730a35a245c22a1c145174e8b8ad31c6d2dbb2","subject":"NetApp – Optimizes the efficiency of pool statistics updates.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77730a35a245c22a1c145174e8b8ad31c6d2dbb2"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 14:24:56.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b3596f79af0cdb442286e6d24642407ed1600d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b3596f79af0cdb442286e6d24642407ed1600d"}]},"branch":"refs/heads/master"},"79c161b8f38fad04030d2a8d11ecf0477f417ca2":{"kind":"REWORK","_number":22,"created":"2026-01-20 15:06:12.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/22"}}},"commit":{"parents":[{"commit":"4fef6d9d4687fa8c1df49d92d3e5ef44fe9711ac","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fef6d9d4687fa8c1df49d92d3e5ef44fe9711ac"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-20 14:47:39.000000000","tz":60},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79c161b8f38fad04030d2a8d11ecf0477f417ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79c161b8f38fad04030d2a8d11ecf0477f417ca2"}]},"branch":"refs/heads/master"},"6077a82b72b1f7607c57c5b915773cadad973357":{"kind":"REWORK","_number":23,"created":"2026-06-10 14:49:10.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/98/926298/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/98/926298/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/98/926298/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/98/926298/23"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:58:34.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-06-10 14:48:59.000000000","tz":120},"subject":"Adopt image encryption standardization","message":"Adopt image encryption standardization\n\nDeprecates \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 metadata attributes for\nencrypted images.\nAdds support for the new image encryption metadata properties\n\u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 for encrypted images created or imported\nby Cinder. The deprecated attributes will still be processed if\npresent but only the new ones are now assigned to encrypted images\ncreated by Cinder.\nIn case of the latter, the new container_format \u0027luks\u0027 is now used.\n\nMigrates the individual binascii.hexlify() passphrase conversion for\nLUKS to an outsourced os-brick library function which now supports\ndifferent secret types, including \u0027passphrase\u0027 for which the hexlify\nstep is skipped to provide interoperability with encrypted images\ncreated by Nova or end users based on a LUKS passphrase stored verbatim\nin the Key Manager. The passphrase conversion behavior and key handling\nfor images created by Cinder itself are retained.\n\nAdds support for converting encrypted images from qcow2+LUKS to LUKS for\nCinder volumes.\n\nImplements\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\nand makes Cinder able to consume LUKS-encrypted images in either bare or\nqcow2 format created by other entities (e.g. end user) directly.\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/926293\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: Iefa3ecc8e68a832f88d7f63ba2734299587d908d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6077a82b72b1f7607c57c5b915773cadad973357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6077a82b72b1f7607c57c5b915773cadad973357"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
