)]}'
{"id":"openstack%2Fcinder~926728","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I9c37ada5e6af1f3c28ebd5c3c2a8baf2d88d0a96","project":"openstack/cinder","branch":"stable/2023.2","topic":"2023.2-hitachi","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-08-29 06:50:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c37ada5e6af1f3c28ebd5c3c2a8baf2d88d0a96","subject":"Hitachi: Fix to use correct pool on secondary storage","status":"MERGED","created":"2024-08-21 11:36:38.000000000","updated":"2024-08-29 06:51:22.000000000","submitted":"2024-08-29 06:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"926728-2023.2-hitachi","meta_rev_id":"55866d12f1c17cef8b5fe9ca5dec75059f1f4ab0","_number":926728,"virtual_id_number":926728,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-29 06:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":1,"date":"2024-08-21 14:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-08-23 15:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 13:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 13:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 11:36:38.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"CC"},{"updated":"2024-08-21 11:36:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-21 11:57:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-21 12:17:16.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-21 12:21:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-21 12:42:49.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-21 12:52:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 14:04:08.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-08-21 14:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-23 12:29:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-08-23 15:50:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-23 18:59:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-28 13:39:20.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"e9fc6f495ab3066af17f10e3cd24b9b70be8680e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-21 11:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3efbfe1e66e79747515c6257cd4332c3101da578","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:36:50.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/341/","accounts_in_message":[],"_revision_number":1},{"id":"09946c7a2087f421805514e0c3c776f37391f4f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:39:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/341/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926728/1/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926728/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"1be16d14bbd55409e7248c3484f930fd1ad490ae","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 11:57:46.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/355/.","accounts_in_message":[],"_revision_number":1},{"id":"0efdffb4065a6a90a20e7377b789d362347a9761","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-21 12:17:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926728 : FAILURE in 32m 18s","accounts_in_message":[],"_revision_number":1},{"id":"573737b6b76ae51017d710d33490696dcc6d401d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-21 12:21:46.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/902b9fffc3e74824957367d6fca0c218\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e06f391bc84a441f807611e4722b5e3e : FAILURE in 6m 33s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df2dcf4312064974bb2d7efe2668ea7f : FAILURE in 6m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7ae2463a6e0549f2a6b458a198eed2d0 : RETRY_LIMIT Ansible setup timeout in 1m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2ec61c6798b4b45975671afc54672ef : FAILURE in 6m 39s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yam... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"39d91245903a45db3bb619f24186e376de1213a9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-21 12:42:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926728 : FAILURE in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7f78e9395131c6739e2d8c2b1b49d7090aa7370b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-21 12:52:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926728 : FAILURE in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0c10c0664a2af912c424ed59941d0fe9bd70e468","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-21 14:04:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790138511355c95afeec286e808b54e16d999893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:10:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0948057635d14509b493c130dac5ce47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faff7e576b8c46bebfc3a45148f27e66 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93cc9c7251c045429409a8432cc7474e : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff8e97dd36945378b23b865a33db6a4 : SUCCESS in 13m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d17a481978be4e99ab334bed4100b577 : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de2394ee26814a67b2233a8e0d00f088 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44b4af4677d45b5a4698f63a0c33ecc : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/42050c96870f4902a947650cc0bbd47b : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3fef80d76194a72a5a9b6374c0ee364 : FAILURE in 1h 26m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12625ce8b39040548ecd5af29f5b31c9 : SUCCESS in 1h 07m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a52eac6de35143ee8fd28de9a7335abb : SUCCESS in 17m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5adf570d93d4d1da21641e5b2e6abc9 : SUCCESS in 15m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5370326529ab423e9da89a9717cae97f : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc79787fd328460aa4e941605850b380 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b925681389d48a2866b7eec5b55a052 : SUCCESS in 5m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dafa1d86b21141d8a991b168bcb748b2 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e667e002a0f4a799c5c478b9ad025ad : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5da8fb8a6d16466b93408179eb780ae5 : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/df957a8b5e20497394892e58418ec010 : SUCCESS in 1h 48m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b580135b9d94915b4e5f2887e07f810 : SUCCESS in 2h 03m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c966be0aa8f448a96a7c86fb6e6ac6f : SUCCESS in 1h 56m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/44f9e7d007134959840a81fdbc7478d5 : SUCCESS in 1h 54m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b0f86067adf4041a23e460c1113958b : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf1fbf53d0804b05a25335ed24abbdcd : SUCCESS in 1h 16m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b22a30b22537452a8a9b29ca11c3d87d : SUCCESS in 39m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df8f9f91886f414f91dd652442ee0ccb : SUCCESS in 1h 25m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c7f031f43cf840919231653d11244004 : SUCCESS in 1h 19m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96eff8f9a0224d7c8c6ab5ad4b964c14 : FAILURE in 1h 39m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed3afd6e5c244f2a8a5ebc33ffa86b24 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"392a2c70b279223e6a2f34678f724c6941e0abdf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 14:12:02.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/940/.","accounts_in_message":[],"_revision_number":1},{"id":"c09ab2c7edf95f2622180966ec5c35f5c6941ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:47:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89002373f324fc3bd615bff9145f612\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbcb52833ef942fea99c10626929ffba : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa386cae316c46dd8a21c42062cd4c79 : FAILURE in 23m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/148f80a142974b5ebb77fb86eab0ef94 : FAILURE in 31m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f1999fa202f4b76963b2a72a458c8bf : FAILURE in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"115b019502adc7e2c61f96d4b84ed9db04b43c36","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 12:29:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1f4a6af898ae21bdbada06eeb6dde08750d49a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 14:43:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a22b8449bcc44d286a0cd179b83e6c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/470afbc3ad874151861964925d191006 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e125a509c1a645838978bac3c1c6b6b7 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b13b0409a9c43f9a20277ec9cd58d88 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cce9395b974d41c7a6974dff2da08d5b : SUCCESS in 11m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64c6081ba1eb47f6bd35f06f2c2bd6c9 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba2157c990e4876b39cc9ec47215dd7 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/635b7977654f472a83c28815173997a3 : SUCCESS in 1h 08m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63d72ed06dc9408ba7c1d9bda5180405 : SUCCESS in 1h 20m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc7e9208a44c4838a843bd8d9d31da06 : SUCCESS in 1h 02m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14a791783ae2452293d28ae692c378e4 : SUCCESS in 21m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70fef87b42484be38babd8207a05ae0a : SUCCESS in 16m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0aea1dde6eca4e1daf1dd6b3c259275c : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/44d7a3f2c8f84bf1b588a7b2d8dc9be8 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b62fca9f55b41d78811d382f6517e62 : SUCCESS in 5m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4d516427e8a407ba1dc917624c38484 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9524dc04113a4f0f8a5944a4ba7f4222 : FAILURE in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1a8f61236594bd7992bde53e83d8d69 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f75112d594e542ec8d6656a29d8d71fe : SUCCESS in 2h 06m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/307e63121f194dd2975d8c150449f30c : SUCCESS in 1h 50m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bfbf91343b5042a9be7282eeca8b863f : FAILURE in 1h 33m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c7e89b0bcc034f0bb61560358564d097 : SUCCESS in 2h 12m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8a456ba11f3b4d02aae4345f07351967 : SUCCESS in 55m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67095ef62ecd406481124dcd53b616a2 : SUCCESS in 1h 28m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/db8f31aaff7d45d5a70410dda8c65f00 : SUCCESS in 28m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dca2cd111d5544dfbd6e0fed65b63fad : SUCCESS in 1h 31m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/24d6b2f400394faca03db6fa59a1f562 : SUCCESS in 1h 33m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f36cdc50dd7f43aeab891ccfc1b1ebee : FAILURE in 1h 18m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ced3e54895974817a251344151e0fddf : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"75aa7d144b1a24038fa926b3e3f1ba919cd6ec49","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 15:44:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c41427d221e7fffc99b17338ff30614f675cd585","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 15:50:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"262b8ddbb103ed6647cea2d84af961dae2e62dcd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 16:09:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2b87bbd08a1bf98e14c80fb2e3df884ac9a0750","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:35:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac27215e80f949f69898047e84b8a2cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/083c17e2365146dca79d561d0a7d205e : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a4f513d50734bee9080a7205b80ad7b : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e215ee2ff4bd45b09179fb40941a33bf : FAILURE in 22m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0555cdd1970b4a88959017481be11573 : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebff7f6738d6bb785796be31cb42b3880fe2bf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 18:05:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cee4eea2835c4c838da9670b37b37da7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d1adb2b8974abe88c325a76d72bd86 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2f1a69ec6be449889d0e350958d0664 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b8d612debea428aaa4e4c96adc1b68b : SUCCESS in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2828b6414764646a67c09a6dffd5f2a : SUCCESS in 9m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c322528ab06445c487da4551100e9351 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0997a2c54a4f24a5e24dd67bd6ce7f : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/ecd085e0be18428da21d9773a8d6d230 : SUCCESS in 51m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96080c8048404795b4601ffe3b6fab86 : SUCCESS in 1h 23m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2423d31e9253429fbedd5ac847bb55c0 : SUCCESS in 45m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7900ccbb89841618fc24ad49f01fccc : SUCCESS in 20m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/afb5706512af400290574f859e4160e8 : SUCCESS in 16m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d77aa1dbc485446c8b2ac4e7cb68ee06 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/92c894408152476fa8996ad9737aae99 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8fb41bad9e344bbebbebe7ed7a5e2dde : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d834667b4cf84f1998ac9bd36a56f81a : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4f9816333584bc080b83bf18cfa72c1 : FAILURE in 30m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62f99cb052144f0ab89d9b15ddce95c4 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/12b6b333fcd947268d30d5e0f8e2d005 : SUCCESS in 2h 05m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a94e05c7e384758858c97bfb8786a19 : SUCCESS in 2h 13m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7d28190c38d94490a95386537e613c3d : SUCCESS in 1h 47m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/99bff24c164f4574bc169fcee34aa124 : SUCCESS in 2h 14m 14s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e2b46063d0a441f6853e89e900885ace : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c4942aa5fe8f4a89b416d3baf9c0b965 : SUCCESS in 1h 20m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eff1aa4f3bfd4c13a18c5afaa2ae937f : SUCCESS in 43m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c50ce9be4e084774aa13ee46e9cac804 : SUCCESS in 1h 44m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/122aeda573ce48f0bb79718f2ba8c4de : SUCCESS in 1h 46m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7763b23d00147c089ace7f11d257508 : SUCCESS in 1h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f5e6b2ddb7e486dbbc80e7e7148575b : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"82c7480873266334ab2daf83ac5db69875d41d64","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-23 18:59:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/926728/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/001d886/ : SUCCESS in 42m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/926728/1/upstream-check/cinder-tempest-plugin-ontap-nfs/4687926/ : SUCCESS in 43m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c7d83ec76d1701f05eb7bd3ebe49beb1fad14d86","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-28 13:39:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5fb4884d71554605fca7db7003c613cde4098b2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:40:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a30d2992ad9ab57fda7561003c0afbb2d7fbbd46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:59:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f84e171c5444afc93cefad9701b1ce6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d8a648120648b1a0bb2d493662620f : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca56c2e7d8b349009a6e6d4d5dde0703 : SUCCESS in 13m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b080b248c2fb4ab3b8fab380a8b0a3a3 : SUCCESS in 11m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/177977bc711e410ea1d71553144f92cf : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936f95d547c6487f9ab753943e764e8a : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e969ce6a0cba4759bd1860cde9974b00 : SUCCESS in 53m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58726fdaf09345508109fc8b22432ffa : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbbb0966bf1f4c0ab3c3e8ac66525b1b : SUCCESS in 1h 04m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/225c785199354dcc9895709788d8de15 : SUCCESS in 20m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ad88eea8d6845eaa19eef8b2540a708 : SUCCESS in 1h 05m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e55e086044684e26b9770c86dcb50ac0 : FAILURE in 17m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fadf022d57824b1ea0bf44caa3a1188a : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":1},{"id":"512d64e3676c99d402535109b25db4ef970f59e4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-29 02:09:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48430a4f14d301752fe1a41f688585f834090407","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 02:39:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d4ec7b01c747e7a8266ea922f1568d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6a3628b9ffc42aaa21dda02b1681552 : FAILURE in 20m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e8dfaf16528415895a7a8b9aeacc673 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78c0e01d2d9841f98119039784cb136a : FAILURE in 25m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df19054f41ec438b95cd5fd150bd2df2 : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7755e0d301ea25b4e87638f3c982362eb0204fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 04:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f8dc23842a416596b4c862595a2a9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fabe7f9d7dd4ff3922c63bf07aac91f : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5901eb894e4466eb0377f3177d4b7cd : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0ca057d22b94ec2802cc52212d38a10 : SUCCESS in 13m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ce785f7a95b4d009439b0fb4f90ef00 : SUCCESS in 12m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/558becac05374970aa351ef0b710fc94 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06de1e0ed0f41d4a15fc1d5bebf328e : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/06713994316c47508387c071c3068360 : SUCCESS in 59m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8bb968edcc847c29f50d00d998c9a21 : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f975c95206a549c293f5d9dfffd0dee7 : SUCCESS in 56m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58782eb1cf984a6bbfac4a0afd91bea9 : SUCCESS in 21m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e52cc982fc3c4887bb60423fb7a28fb6 : SUCCESS in 16m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c547bef01e43477f91c8c972904a36e9 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a384513abe134816bbd701211005a6e9 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ccca484f32e4c12abbbd7a631a4983b : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d4d35fb0a5b46dc84d16596d061ed8a : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/274ecfd2ecd3494f925f67e9565b00c6 : SUCCESS in 42m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3dabcb3887d64431af1a755e97b2aa7e : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8208ad816acc4bf6a7c10cc6a67ba1ee : SUCCESS in 1h 53m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c9e464b4ca0541e0b1abb21bdd5e6d29 : FAILURE in 2h 07m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47207e7849fa41089e6b430ca76450c6 : SUCCESS in 1h 48m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ae995f137eb24e46b142992962ec0b28 : SUCCESS in 2h 03m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/435ac84b51e54e85a72b056a26ff4a0d : SUCCESS in 1h 09m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3fcc45d828b14060bfc6a41cc26757f9 : SUCCESS in 1h 30m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/22ef73be37c3418d95234d1be37d8423 : SUCCESS in 44m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19cffa6fe29646a884bc26f4058a7b9c : SUCCESS in 1h 11m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6502bb42cc6841149f4ac14cd411d1aa : SUCCESS in 1h 24m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f344ac302694e7f8a39b0197ab6587e : SUCCESS in 1h 52m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35c76aa6264641a38efb7a6cdc3c0a90 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d6dd8308b27259b4bca97836d70dd7b736abdb83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 04:55:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ad3ac601ed77c21a09221039715048a84b3fd8b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 05:26:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/926728/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/2a82b7f/ : RETRY_LIMIT in 3m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/926728/1/upstream-check/cinder-tempest-plugin-ontap-nfs/20252f1/ : RETRY_LIMIT in 4m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b62552b50a493eed3683c4881b2add572976a542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 06:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e924fd786a41049e0284207a0aa434\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921c70d8d05c49b0875283e874ddb5cb : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/008190417a5c48aab866d9df0c82ec54 : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d57278affa9a4fad834815de5c64785e : SUCCESS in 15m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be8c5aedd30344aa94e45813464f100b : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67a7d850c474762b2f2174111cad5f9 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a762ca6a03014e1cbdcf3f1f3e9ff607 : SUCCESS in 43m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b419ad7d54a4f258eb8d7bfa263f974 : SUCCESS in 1h 27m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48726a36c2ae4babb584448a655f1e3d : SUCCESS in 56m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eee5a683ee7647d2b67be1157a0a9fbf : SUCCESS in 17m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4cfd88135ab44e73a47cf538ad3ab429 : SUCCESS in 49m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a151e089afa4f2d86daba1f224f927b : SUCCESS in 1h 48m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa21f092c4414547ad78b50260885a78 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2a2d6df3671ef345c719341e733b1e63ccbc6a2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 06:50:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55866d12f1c17cef8b5fe9ca5dec75059f1f4ab0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 06:51:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec04be50794440f959d31b2a9e21627\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff46913ebb2c4306b91d232e2515cb4f : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a82d474821764b27aedfe5d0dc522f57 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19540c1aa6e2b39e6d47e738037b880ede21be73","revisions":{"19540c1aa6e2b39e6d47e738037b880ede21be73":{"kind":"REWORK","_number":1,"created":"2024-08-21 11:36:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/28/926728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/926728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/926728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/926728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/926728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/926728/1"}}},"commit":{"parents":[{"commit":"9e667b02b2c20b4ada18c1a472be152956284d45","subject":"CVE-2024-32498: Check for external qcow2 data file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9e667b02b2c20b4ada18c1a472be152956284d45"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-03-17 05:52:44.000000000","tz":0},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-08-21 11:13:59.000000000","tz":120},"subject":"Hitachi: Fix to use correct pool on secondary storage","message":"Hitachi: Fix to use correct pool on secondary storage\n\nThis patch fixes to use correct pool number for a secondary storage on\nGAD environment[*1].\nEven though the option ``hitachi_mirror_pool`` is to set pool number for\nsecondary storage, the option does not work and a pool number for\nprimary storage would be used on secondary storage.\nThe bug should be fixed and delete the risk to be used unexpected pool\nnumber.\n\n[1] GAD is the storage mirroring product on Hitachi storage. The feature\nwas implied as the patch\n https://review.opendev.org/c/openstack/cinder/+/796170 and was merged\ninto Antelope.\n\nCloses-Bug: #2011810\nChange-Id: I9c37ada5e6af1f3c28ebd5c3c2a8baf2d88d0a96\n(cherry picked from commit 2270611507720ba5c05bb6acaf0a5825711ba992)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19540c1aa6e2b39e6d47e738037b880ede21be73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19540c1aa6e2b39e6d47e738037b880ede21be73"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
