)]}'
{"id":"openstack%2Fcinder~926731","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I033a76b9e04b2032c38b8b1f087033409470fafa","project":"openstack/cinder","branch":"stable/2023.2","topic":"2023.2-hitachi","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-09-14 03:11:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I033a76b9e04b2032c38b8b1f087033409470fafa","subject":"Tests: Save 30s on hbsd FC tests","status":"MERGED","created":"2024-08-21 11:36:38.000000000","updated":"2024-09-14 03:12:46.000000000","submitted":"2024-09-14 03:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926731-2023.2-hitachi","meta_rev_id":"b51e20e77607ae5de9313b9207b7ed31fc3c8512","_number":926731,"virtual_id_number":926731,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-14 03:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":1,"date":"2024-08-21 14:07:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-08-23 15:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 13:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 13:35:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 11:36:38.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-08-21 11:44:10.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-21 12:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 12:31:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-21 12:45:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-21 14:04:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 14:07:57.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-08-21 14:09:39.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-21 20:48:02.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-23 15:54:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-28 13:35:49.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"8ac894313f2f2d48b078990f04ddbc21e7a0758f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-21 11:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2195e766bbb7b02020496132682049fd1a03e8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:44:10.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/344/","accounts_in_message":[],"_revision_number":1},{"id":"0d7acd5ee06e3e4c4ac73b55bbfa2305a8acb7e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:46:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/344/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926731/1/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926731/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"97d613639d9f21a6e98368e72f78d540e40f777a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 12:25:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700de1ee369349f984deb27287814568\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e82d1093c74473abcb4e370ba789da : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9c79b9c396f433abb8f1fa32f74fcb2 : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0daeff2da7f94f5cbbeac4299fb2eee2 : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ceedfc11b5fa462d84baf58a8efcdd31 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34cda7f44cdd43ed9863672b2831bb3e : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7817edf03c546dd83df6254728f2ebf : SUCCESS in 16m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/71ad447933494cc7bc1fed545ca7ac1b : SUCCESS in 18m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15b5728e41004bdfbc11e44f07fadb59 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3b75b6fffe4c5d2c0199883f868b53cc13b3236c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 12:31:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_12_25_36\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"d164a3afa6efd99f8cc810c653e5790eb85fda36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 12:45:42.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/358/.","accounts_in_message":[],"_revision_number":1},{"id":"c1e0d3592abbce637166231d9a62837634a3b8fb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-21 14:04:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d863229a1dd304c28d3c78104abb4d65f0bd280c","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-21 14:07:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"694b1ec5d088d18737d76375135d7052268af812","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-21 14:09:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926731 : FAILURE in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6efa6bdaa3b970c11736ba37e9e038445a950503","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 15:32:46.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/943/.","accounts_in_message":[],"_revision_number":1},{"id":"af7be346ecadfa6bbc61f66ae9eca1959d7e3268","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-21 20:48:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926731 : FAILURE in 37m 41s","accounts_in_message":[],"_revision_number":1},{"id":"984c63855f786e543fd8868f59e3437e30985ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:02:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847362f89aac4f31b20365b907186b80\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/add55ec6fd0e4d4e8005002e7b344dbf : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5b08d9bf80a4a76af1e9ecb448bb849 : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73357be118354837a84235bdc3904bbf : FAILURE in 24m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22b949f5c7cd494ba10db9b362823bb3 : SUCCESS in 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aa44f1e330b7ce28f46016c88b39689a91d7339","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 15:54:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"781ed21cd8644328644c163c8dec65438a9797db","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-28 13:35:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f66bb4a148180c5ba4df4dbc1048a79127a86bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:53:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8356479c4addd8c2a3b1bd06f3e26c9b42ea526e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:00:02.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926730 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"07ea48705a0174716f6782bcade12ca6e2473434","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-29 18:10:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2424574c712e1bb0d7d8de3211e67990a3bee884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:32:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fd23cf45f84c5fb16a5592fac6effa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ca70f5f5fb4cac9166ad10c1b8f904 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/106b00c3e4a54a399e4c776941c99236 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ca5bdbd92bf4783855c4d2fc2206b6b : SUCCESS in 10m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a9edb042a0342f5b1dbf64dfaa53eed : SUCCESS in 12m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d379cec65e7e4293a8e4417c01fa62ee : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486c3190ea5d4531a9927c453d9816a2 : SUCCESS in 14m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b69696a452974bab8d0b092ed122f40f : SUCCESS in 14m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/629e27750f7d48efbc64397fa66f1a75 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"828910ab9a046f72ea98720032c8c780459c8f8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 19:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2343bcef03b548f8aa4195c883cb9bb6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/448b2c5cc4864dbfbdd1ad0a50fb3eff : FAILURE in 20m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65efd2433c8c44288e22de030ed20071 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c423727abd714d1f996152265a7bf45e : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/27385b7b4a054cd0ae9d56018dbe2301 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec9b7f7e264b03f974768bdd7b3de6c03ae0c0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 15:23:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"942472d5469f4c926ac8703626f9340cf5e439bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 19:48:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926730 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"04353190b2f34adca1c300352bd6e31140c23b52","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-14 02:27:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e8f040b6139e99986510cf23b362f7a88aa25a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 02:53:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/548717a06ab64dfd8f7cccc314517b85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8faff80c7b5e488aa904a0f88cedc1c5 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25f09dca44894b979e20dabc13e5c071 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e225dfb3e1404ea1ee5b52553a1b1b : SUCCESS in 10m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbac17ced14f4f398dd948193901b721 : SUCCESS in 10m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0c2f19af3e447e4a36fe6fa0fd3a285 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6ab8fa401544a69af946dcee692044 : SUCCESS in 12m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2715c86fd4c04a639ead5f6c1db54ed3 : SUCCESS in 16m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ea2139a63835434986bffb8ffb33aa07 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"33ae079d9a904adfc348b6796a768202a52d31a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 02:53:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89b9ff6ee4751689e4445ef2fba7f095d3fc5542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 02:55:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6c4a5d09fe1487880d11951aa503351\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37d98cdee79940499ff7604d03898d34 : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b96ba031ff347528c3d3c91cfea6c64 : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59c305b6453046d4a1d91169bcd98345 : FAILURE in 18m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/27861e456fbe425f8f184d1b6e75836f : SUCCESS in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"205077ef44d2e524393ea0b70303d02d076c158e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-14 02:58:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_09_2024_02_53_33\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"cd50293bb30f6f30387aa0df8058268c28d50675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 03:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a7124f119d48fe885b29d2fcbda8ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99fe8d6297db465ca4034da7847f7944 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f391abb3a80d42d7a16623f0ab4be141 : SUCCESS in 11m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efec107f83c14cd2a80e5df30385c46c : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9de391dac922418487de3100fab3d92b : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b89a13e166f42509be89c366003e7d0 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8840e41f59f678034714d5c9b8949b86a71e6e5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 03:11:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b51e20e77607ae5de9313b9207b7ed31fc3c8512","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 03:12:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf73867301248929d5a5b0c17515310\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429c1f9534454859a9dd2bc64824a515 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6137b9653bc0f722aabf68241dfaa1cf0116c08e","revisions":{"6137b9653bc0f722aabf68241dfaa1cf0116c08e":{"kind":"REWORK","_number":1,"created":"2024-08-21 11:36:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/31/926731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/926731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/926731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/926731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/926731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/926731/1"}}},"commit":{"parents":[{"commit":"8f497c1cb3e0ebf524c325ce9be365e76763da71","subject":"Hitachi: Fix exception when deleted volume is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f497c1cb3e0ebf524c325ce9be365e76763da71"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-30 19:37:12.000000000","tz":-240},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-08-21 11:14:26.000000000","tz":120},"subject":"Tests: Save 30s on hbsd FC tests","message":"Tests: Save 30s on hbsd FC tests\n\nSaves 30s of unit test run time.\n\nChange-Id: I033a76b9e04b2032c38b8b1f087033409470fafa\n(cherry picked from commit aa91cf149e73201df1511e487b0ce0227e834912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6137b9653bc0f722aabf68241dfaa1cf0116c08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6137b9653bc0f722aabf68241dfaa1cf0116c08e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
