)]}'
{"id":"openstack%2Fcinder~926732","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I81090aee4ed6c288f7d9bbdb45d7cc849663e393","project":"openstack/cinder","branch":"stable/2023.2","topic":"2023.2-hitachi","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-09-14 14:51:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81090aee4ed6c288f7d9bbdb45d7cc849663e393","subject":"Hitachi: Stop frequently REST API request in test","status":"MERGED","created":"2024-08-21 11:36:38.000000000","updated":"2024-09-14 15:25:58.000000000","submitted":"2024-09-14 14:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"926732-2023.2-hitachi","meta_rev_id":"84f229526e2047b9ef2dd8fdada221fe806f7038","_number":926732,"virtual_id_number":926732,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-14 14:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":1,"date":"2024-08-21 13:55:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-08-27 12:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 13:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 13:57:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 11:36:38.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"CC"},{"updated":"2024-08-21 11:46:53.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-21 12:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 12:22:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-21 12:38:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-21 13:02:01.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-21 13:55:14.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-08-21 14:07:10.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-21 14:22:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 14:47:10.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-23 15:56:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-28 13:57:05.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"acd9ae11d390e9433ee4c128dfb8814b3bcb77aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-21 11:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5e08f5a3db895e0adc25fb1eeab76512106ff2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:46:53.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/345/","accounts_in_message":[],"_revision_number":1},{"id":"00ae52365983755d46735b1b151228b3907fa902","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:49:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/345/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926732/1/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926732/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"81fa9939eed6d7cb18cbbfd165dd3f4395c6f3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 12:17:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0e47164f2f84361bac6362973d47571\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79e074e06e0408ba264f669d9c5dc51 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9331645dfc44ae6a28945100b3068e4 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/007cee175480493d941d169d0fdb5eaf : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aabd760f9f814f07a290948955bad315 : SUCCESS in 9m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74d6a448723b449aa0df3791f7868a48 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2efbb181e644f3a93196d38a62335b : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e4a521cf87140c295c6e6e47b99e69b : SUCCESS in 20m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c130fc93a76042a4b9ef6c9191cbf1c9 : SUCCESS in 18m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab1b01bad5374be48879f1134805dcfe : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"929f8cadf0fc9c13de5032d7518761cbeae50a07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 12:22:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_12_17_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e6830c3d9f34f289df0d9372ac1fc25f80cbbd7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 12:38:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/a59732a/ : RETRY_LIMIT in 8m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-nfs/c192be3/ : RETRY_LIMIT in 13m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e5cb84cbb24818f8e08a12fd860d9ff133d475e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 13:02:01.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/359/.","accounts_in_message":[],"_revision_number":1},{"id":"8c879f1b9f946207e7a4fd2ada87c8bc08bbebed","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-21 13:55:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"910c81ae13ebe8c149d0538eb4cf73148573e0d6","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-21 13:56:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6285bc5d0e0d7ba33c61578043d8304bf828fb24","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-21 14:07:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926732 : FAILURE in 36m 09s","accounts_in_message":[],"_revision_number":1},{"id":"83d2103c44b56b84ee0f96c020f74034c650d0af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-21 14:22:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9b8a62ce7937c75919e877164547e39d38c2db02","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-21 14:47:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926732 : FAILURE in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c4a442a3e1a60e08bb54a7729fd72f4d9aabda1c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 15:59:44.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/944/.","accounts_in_message":[],"_revision_number":1},{"id":"99850a578a69d52ff5c574139723535ea82670b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:43:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b176db9da4d4640941fc853ee6d8e2d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cbbaa9c297b40e7b0b1464ec030413d : FAILURE in 23m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/932e8d781d6f4f69a4f4ff1bc194fddb : FAILURE in 30m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a10b957b150347d7a100bb61eb259cde : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9688d05ccc6f4da9bf5ebab9b39aaee2 : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac05bf8d2c050e7932d8f27ad8b57d0f3b0ceaa2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 22:05:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/a6808a0/ : RETRY_LIMIT in 8m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-nfs/efd08e3/ : RETRY_LIMIT in 13m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b7c7ece258bc86a884fe52f3f2186f79cb88526","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 15:56:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7221ea6095db60031b7bdd0fdbab1b9b7b25a385","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 16:17:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7336e2b3f3a698d3d47e9a37d01846c55d5e89b8","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-26 09:08:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc1cb8c98ad89d0aedb9da6194b489e1056fb7e5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-27 12:09:15.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bece360e57dc6e2ab8d635dfc9bd251b281c15d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-27 12:35:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b47b0b7b611ff6646fee3924f2518c4587c7b997","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-28 13:57:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"089278625b5ed2a7bcbd40c295cc2fb8fa129f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:57:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68ccf5a477203b604fc85deb803dbf72e9a46347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:00:02.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926731 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"c9705fc78cb40c46f999d0a09d8a65d2ecaa3620","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-29 18:10:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"202bad03c2cc42929bb82905b5b8dd18f582836c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:32:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a2c85581a3540369f20f8a0d7cb2cff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b991b1756d7f45f9982c598bcf76892a : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f2f4ad5df2747eba16f5e8ea23dfa55 : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aae8666fecc48b6bb39bb0841ac337d : SUCCESS in 11m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6d77121ddb04f59943974a9918909e9 : SUCCESS in 10m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d23ec858264f4b0e87ba6655e705eeb9 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3ff197e9894a9880fa882dd170a9dc : SUCCESS in 15m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d115c8f12a742c5b1013bd25395292b : SUCCESS in 18m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4dbb843ca16b41b28c25bbadfac7e2ef : SUCCESS in 16m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db6c947078154378abd006963bdc273f : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"443025f6daa05f5a8f372dbb3dee9c5196ba66d9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 18:48:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f115142/ : RETRY_LIMIT in 3m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-nfs/71fad86/ : RETRY_LIMIT in 4m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2fb4869f238112a7c9fa78ebf1acbeba0061003","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 19:16:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0698d20c3a04caf877bc5921e066b53\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34bb5f2e7b13408a9a6106e397b13f2d : FAILURE in 25m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b8685081b684ecbb9ac0bf5ff853454 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/653b3844addf450e9264ddbb473c4752 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f868f1f9c7d74801bbfa244488b1d5bd : FAILURE in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cd01ff7c8b898ca7dde592ce2becbf0d8ad0066","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 20:18:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/1b90e99/ : SUCCESS in 45m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-nfs/800753a/ : RETRY_LIMIT in 3m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8e4e422a5941e7549bb6910c540dd37e6b7969e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 15:23:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b330829b62d45c7b530718e777050aabc82d4999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 19:48:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926731 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"136e240147a936fb089ddeaeaba712d171a1e3db","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-14 14:09:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0c2289a1e15f81f2d48a7a662ae78aef75b373f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:32:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9439738ec424ad7b5649830a362a5a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70169868a7504a95bdda2b4447d1a0ce : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8770bdc646ea4e4dae3ac4487ebbe5fc : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59a160f8aa1a4cb7ace6ff2076bd89cf : SUCCESS in 9m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d060606163cc4ba6a77d3fe7ca5d998f : SUCCESS in 9m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f79f132cb2aa4afabe69b85b5f7bfcd1 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0179c23b224491a93cce1b9e013f453 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2069404a3fe44fa6b9298ad6451f7307 : SUCCESS in 22m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a351149989244932ac015fc71a6383d7 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e09cad4a87a4e668913a988608a342c : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a3acda8374ebf34d1fb5bbbfa8cf346568b75fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:32:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e8e965bd4fe3eb3faf28999ab2fb3681afd6aa2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-14 14:37:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_09_2024_14_32_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9ec1909bb5ae6616a47a53bff11c35c92336b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:51:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43872f89d40d45259e3b4fece1ffd0a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4fa20f8f0d44fc7921f18fc5deeca39 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe4ef9ef422413f856368b4bf4566d1 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ca136e4f4834476bb2102f060347c98 : SUCCESS in 10m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af4bf73888449f2a9f1af1632542c8c : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c7477aeb694cce86d4f7165899a3b3 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/046647ad90ec415aa35e3c6784d75e0d : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4d3b91c087b9354b62e2bdcc1973a2703636da99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:51:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8060c9887f15ebeadfe548f0404147e7d6f314e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:52:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519a8b4ef0e44509baa58e29b79b429c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e252d476d754422998813ec4b468895 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffc4707f9aad4538b6be9ee23a4a3b28 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"ca2658138b40833751d351946b86392539ada687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 14:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173edd91e50449cbabe73d1fc43ed835\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/632601291aac40c88bd2753bcd9b4ef1 : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b5b14eb601b4821862d7b68eb6cbd6b : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5de0734b5e945c0b2410821d6ca4c7a : FAILURE in 25m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4649ca8b43414fdb92eeec7a9416df9e : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84f229526e2047b9ef2dd8fdada221fe806f7038","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-14 15:25:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/811c932/ : SUCCESS in 45m 59s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/926732/1/upstream-check/cinder-tempest-plugin-ontap-nfs/106a28d/ : SUCCESS in 41m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"070ca24931e36154700119199dbf5f1606302981","revisions":{"070ca24931e36154700119199dbf5f1606302981":{"kind":"REWORK","_number":1,"created":"2024-08-21 11:36:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/32/926732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/926732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/926732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/926732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/926732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/926732/1"}}},"commit":{"parents":[{"commit":"6137b9653bc0f722aabf68241dfaa1cf0116c08e","subject":"Tests: Save 30s on hbsd FC tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6137b9653bc0f722aabf68241dfaa1cf0116c08e"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-04-25 01:21:48.000000000","tz":540},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-08-21 11:15:15.000000000","tz":120},"subject":"Hitachi: Stop frequently REST API request in test","message":"Hitachi: Stop frequently REST API request in test\n\nThis patch stops submitting frequently REST API request in test scripts,\nto avoid a risk of failing the scripts by unexpected REST API response\nfrom a psuedo REST API in the scripts.\n\nHitachi driver submits a REST API request frequently, to avoid REST API\nsession timeout.\nit should be stopped while running test scripts, or the request bothers\nREST APIs which is for cinder features, like creating volume and delete\nsnapshots.\n\nThe test scripts have codes to stop submitting, but one of the code does\nnot work by using incorrect variable.\nThe patch fix the variable name to stop that correctoly.\n\nCloses-Bug: #2063317\nChange-Id: I81090aee4ed6c288f7d9bbdb45d7cc849663e393\n(cherry picked from commit 60b7062902274de084e8fefb4435b094b92be894)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/070ca24931e36154700119199dbf5f1606302981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/070ca24931e36154700119199dbf5f1606302981"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
