)]}'
{"id":"openstack%2Fcinder~926733","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I2a2cfb44207b03fbb006d29ae958896f3f3231ed","project":"openstack/cinder","branch":"stable/2023.2","topic":"2023.2-hitachi","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-09-14 18:05:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a2cfb44207b03fbb006d29ae958896f3f3231ed","subject":"Hitachi: Fix to set correct object ID for LDEV nickname","status":"MERGED","created":"2024-08-21 11:36:38.000000000","updated":"2024-09-14 18:06:20.000000000","submitted":"2024-09-14 18:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"926733-2023.2-hitachi","meta_rev_id":"a438bfdbcfe7f6b72010de436df4674ce1b70a52","_number":926733,"virtual_id_number":926733,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-14 18:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":1,"date":"2024-08-21 13:53:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-08-23 16:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 13:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 13:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 11:36:38.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"CC"},{"updated":"2024-08-21 11:49:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-21 12:24:41.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-21 13:15:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 13:18:49.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-21 13:53:56.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-08-21 14:39:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 14:42:17.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-21 14:45:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 15:26:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-23 12:31:31.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-08-23 16:03:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-23 19:42:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-28 13:58:34.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-09-14 16:29:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"0c75f3efa04240f3a66e8880a58c87805e00f766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-21 11:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82a14c73f183d6768d15530b91ecf78fbd83d3c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:49:50.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/346/","accounts_in_message":[],"_revision_number":1},{"id":"ae4b8e4f9661c6f0fbbd32e7e64ee7ae8ee9aeff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-21 11:52:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/346/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926733/1/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926733/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"edc634c98090af9445372df008742e8eb7e5feeb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-21 12:24:41.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/256e4424db4649c39448d1d0cbee1051\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e700be9c6a77437a95a1cf04bfe2a1c7 : FAILURE in 6m 32s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d6710b151b2477395e87ebad76a7e66 : FAILURE in 6m 53s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c0df577b98b34f408f6b5620e16073bd : RETRY_LIMIT Ansible setup timeout in 1m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b1d1be427eb4b288b8100bcbb767ea0 : FAILURE in 6m 54s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yam... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"93cb47e358392811aaf416fa82c9c0fbad827acf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:15:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d1a4b454664c679195bf20efce9932\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbd89c3d0cca467c9f4fbec6f4c34558 : SUCCESS in 16m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/365e55a235eb4588bae79c28279c0847 : SUCCESS in 16m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eebd9e6f90784eaaaa3a1a0f950547e0 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1f60c31f8634ab1b71504110ab01047 : FAILURE in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9388b813bd220b845e8ad862dab87204ccefd90","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 13:18:49.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/360/.","accounts_in_message":[],"_revision_number":1},{"id":"4f35912b47a64483e7bf3d751871673fac0c378b","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-21 13:53:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbef13e06837a42cd3050d7eeb2a1606f63a07c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-21 14:39:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f625901981045d4f87429576d5fe61e4be1176bf","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-21 14:42:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926733 : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5731f88c5d23353ccd12dd29508dd4e33509e054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:45:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea133b63c7504038998003143272d27a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9a5def3c76499789d28558713bdd44 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72ca952ee5c343729e045ba86af75bd5 : SUCCESS in 14m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65e0b6848459484ab6eb3ee1a5f54222 : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f57159a5bff840408d13ae095bcab283 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/322fee0c474241bab947191d496469a0 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed883d13f4a4445baa73c3a5ad9c700 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/64fc507e642c409d8107f21bf644740b : SUCCESS in 1h 09m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3789d5eb8d04c95aa5024154f03f421 : FAILURE in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df5a0c85e4a744799f357491f40ccdc5 : SUCCESS in 1h 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd6d021375404009b96d4d516240d68e : SUCCESS in 20m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/998b99e871374a03be17b826376f073c : SUCCESS in 16m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed6f220a4ae44777924c9b50e1d01af5 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93aa7847fbe24b4798afa63441b93e36 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29887f20d6e74128925009332c19f91b : SUCCESS in 6m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed4e04da9b9b48df875c8dad8d8b7aac : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e809d6d309c04a23a44ae58b4e71ffa9 : SUCCESS in 37m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a7d4409ad28447593ebd1fed92aeae6 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7023d3627c444bf68122a1461e14cc3c : SUCCESS in 2h 16m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8848b774c1ef4e5f9c64ce810070525e : FAILURE in 2h 35m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b3df1f56231a4788be33ecc2a608c04b : SUCCESS in 2h 10m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7b75255931644eb7b9b0b3b76c3e10db : SUCCESS in 2h 08m 47s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/436224eac1634448b9b188387fc86ae2 : SUCCESS in 1h 01m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3a1c5a156df42ceace8c52920923a9a : SUCCESS in 1h 46m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e976457dfb4145019b1a82368f13f98a : SUCCESS in 37m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d18ae36f6928442284bf434a72ad3de9 : SUCCESS in 1h 29m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a75d79f0a575476394a40d8e854b0c34 : SUCCESS in 1h 37m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ad91246804c4990938a2b5d4d736ead : SUCCESS in 2h 09m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf396223a104a7292333f6a72bea565 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"589cc1134f4f6ab9053ab6b8ad81a8b2f1505486","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-21 15:26:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926733 : FAILURE in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dace81220bf67dd4366a617c1bd37dec3df12f16","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 16:26:07.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/945/.","accounts_in_message":[],"_revision_number":1},{"id":"af7b931a63976ef656ef7ff7c21e2d03d33b1740","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-22 08:53:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a29d1a193dd425927e21fd339fad365f078f9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:01:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6255531e72f94919aadff13a57cd68c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7d35ec6d464ae3ad4fea82f4513e97 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750a3216fad6444c946c1e22f21e09e9 : SUCCESS in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b262b697b8a47dc88ea495acf383b4e : SUCCESS in 14m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e19357986f8410ea13091f73a735508 : SUCCESS in 11m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c004d13dc4e4ea68312f9b1e85fb075 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bad5156a0d42d99708a4553fef0ab1 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/64fa83ea339e4a6cb23bbbcd5b393a2c : SUCCESS in 55m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00c07d7f5ac84d28983ce500415cf5bf : FAILURE in 1h 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f12a9a456d7e4fe696e9b57fb0ec1c04 : SUCCESS in 55m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6308ee42b4f74fd587925cd3c8567445 : SUCCESS in 20m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e6880038413f4ca3896c230010fc8a57 : SUCCESS in 18m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f28435f433974beca8f5e8ec5877fa10 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b42671b3fb840219f7f9fa632347f48 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6378216e7d774dd586faaade91e6e9b6 : SUCCESS in 10m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0943a391327246dab75cdc7c44b838be : SUCCESS in 10m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/45aeaaaea6f9471593a28432151790f9 : SUCCESS in 42m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98daaf7353914ae68370f9b09384c229 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/00829f2237574a69abc09f1b41d668b5 : SUCCESS in 1h 44m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50f6a754e01448579b7e4a9a41967080 : FAILURE in 1h 59m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5233f5d6b7104723bdc0b8fbf96e831d : SUCCESS in 1h 48m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9cd74c2a59d443118688edf57aa57f68 : SUCCESS in 1h 41m 29s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f43465c1da5410b808bb17ebca9f69c : SUCCESS in 1h 11m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/117116a6f7684436bea4eeaa69967ce2 : SUCCESS in 1h 22m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/45d4b117cb214360aae9bd44024e5959 : SUCCESS in 36m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/542ee64c398b4c7491b23b3e2cfa2cb9 : SUCCESS in 1h 13m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3278da0992f64e0699244f478d39edac : SUCCESS in 1h 32m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5bb69ad8ed842449f31edd48676bad6 : SUCCESS in 1h 25m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f71eb3c9675443bbafab0e47284dd9f : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a23681613f854621702111756b25cb10151b3379","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:09:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdcca531d2e94c61a27ef6d30ad7fc98\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4c081ffcb494d16a8ea1ffa3527fb1e : FAILURE in 33m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dfb0fd0a8b5435f8b5c5c94a24cfe2c : FAILURE in 27m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dac29bf5fc9b4b9c9a39a0e2e678756d : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1798188bc3dc43adb9de85a18cd48749 : FAILURE in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10aee22f63ecc33189c27d75b9420c1fb2247852","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 12:31:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93d8316d6dc6e70bab60d2a95545096d8eb92400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 15:03:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/046a27aeefb84664bb12f0f267b4074f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705978a16871429ab0e593d845eca044 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5f83a1cecdb41419768b79d8ddf5dd3 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c829d6b4ba2b4099b116a9d177bc6e29 : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0c5d43bc6324acdaec5d3a2f2cc244b : SUCCESS in 9m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ece2142ce4d64aabb5115fa55151fca0 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764251efca25450f87ce4962b887c47f : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a27e0f71fabd451fafebc2203e1e6430 : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/467ef3455cf644e8937bd36976e13921 : FAILURE in 1h 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b78f3dee9ea4491898d9197562d2a058 : SUCCESS in 51m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c751ef017d475f949e2faabb72bd77 : SUCCESS in 20m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf44734b923f44f18dbec0cf40f3036b : SUCCESS in 15m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f667063b500e40148a20febaf19dea68 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b2f6e14dc41487780c74c7e0f1f685d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65267eaca3da4d098e04220964235d0b : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07acbcb467e749e6a09ad1656caee5ca : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f4f47b64fe2f4f8e89b45adf5ad5f368 : FAILURE in 29m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62e46b6425534314991b5d7a9e49e473 : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e138e2e0396d4d25b5310d3c7530bf1e : SUCCESS in 1h 50m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/46fc897590184acdba73749dbf4438b7 : FAILURE in 2h 26m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8375cefe54f64e869284b9e6e5f85072 : FAILURE in 2h 04m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/59baaf5ff4e84b8286528fa8a989efb4 : SUCCESS in 2h 10m 46s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e98c536bf1f54fb39a80da099172f9f0 : SUCCESS in 59m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fec8cbefef6e43a5916cd8eb16599928 : SUCCESS in 1h 26m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5c656b71e6aa409ab9b0cdb87b91aef6 : SUCCESS in 39m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06d8774783024a99b859c0a6007a909a : SUCCESS in 1h 16m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cf628fde442b452d98a55a4a4f755929 : POST_FAILURE in 1h 49m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac66c77c7366446ea8802b05bed07e43 : SUCCESS in 1h 42m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2bbb33e7d474d3591f9aee67bdb039c : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"47c51830ab512c525943c2fb4e336397ef5bb877","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 15:58:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db50b115f8b06417d08c0851aee118953f2c36b8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-23 16:03:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"254114637dcd5dcb1566c015f65c3bc96dafb491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 18:17:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba0b5ebeaab4fe1a512ee92b053642a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90ecf78fe8b74c7089ac1e971756255e : FAILURE in 20m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/760063363b534659ab83e13dc203ea54 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83fe024a910b4927948f952e8311b381 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b6e1d2b85dc148db9d2a49be6bbaf4b7 : SUCCESS in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02d7731e06eb1c1181d20a1e608745c873d87fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 19:00:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb0805c4beb4f51a81c7bb999e6dd0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d595ddcc3c40f8a97bab6d0ed9b1a0 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0c39d14643d4cdba657d182cef1c1fd : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec41a326f94c490ea4b55e99e7e3d855 : SUCCESS in 12m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18628276b2d346ada5a823643e8287cd : SUCCESS in 11m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3705d1943184af6b1db93a0e755213e : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12089d1d0ee43a480c6be9e80f165f0 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/1e72ecb7206e47e1992a69befae90900 : SUCCESS in 42m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6705a552d1d344f0be6cdcf873af1be6 : SUCCESS in 1h 02m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ff8f5461f0243ca8135f45cf4578b31 : SUCCESS in 1h 05m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ac7ee6247d4014b117639bfe3889b3 : SUCCESS in 21m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de49012fb05c465ca4907fa5bbe72b29 : SUCCESS in 16m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ea668bad29e4b1088e66175f3770fc5 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b5548ad1feb4a97aa9c3a284ea5bedf : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94ee13410d86419a87b059c2174bd7be : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6930152afb9d44c78d17310cece8bd10 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e4ababcc1de749b2ae1f70a71691510d : FAILURE in 46m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fee17405225341c5b5be2a8da4956cdc : SUCCESS in 5m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f209aae25d6944e0b04ed7bfd1f5b7c0 : SUCCESS in 1h 43m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd89a8543fd440a69cb6fcdc9cfc0b1e : FAILURE in 1h 41m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9341e2ecbece4dab8da2cfd9f4778cc6 : SUCCESS in 2h 53m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f6a8255362c040149b1a3011b2c714f2 : SUCCESS in 1h 56m 47s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce65f9fa28964e4099b0c782cb8e5821 : SUCCESS in 1h 12m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/725e0598441d40529482de0f3ec60809 : SUCCESS in 1h 29m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cc59559af4654db4ba4369688b223ad2 : SUCCESS in 43m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddc12e7781f84602adf19968c35d2fb7 : SUCCESS in 1h 12m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4aabeedcca7146348a06cd0be3c5cef0 : SUCCESS in 1h 22m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85e342dddf944df0a1dd8984c4a46e6f : SUCCESS in 1h 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8664e4f64574ce39df0acad76a75ef1 : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1},{"id":"338f6ef764a9404a11e210045694139072451e69","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-23 19:42:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/926733/1/upstream-check/cinder-tempest-plugin-ontap-nfs/4c26345/ : RETRY_LIMIT in 9m 22s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2edad06c43caea49734d78dcf04b54fdd1d9618d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-28 13:58:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"695c50269882745dbc26fa484d3142d96b62c8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:59:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ced1258a58fc3ad1019b4f3c3264419b6a45a608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:00:02.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926732 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"cae56c88dfba5bb014c93b46c92c871b36052aa2","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-29 18:11:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ec5edb83ae9a0d5874002aa498a908abbbb6f1b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-29 18:20:02.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ee61e1cc99a6446fb6ad76eedfeb93b4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/29cef40bab0c40808aee30059f67d7e2 : FAILURE in 6m 54s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/09d41865dc71484ea251343d89399e3b : FAILURE in 6m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a8fb2cd190934cc1923d0eea5ff9b4e5 : FAILURE in 7m 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4c567f3b4af64bf0bce4d683eb6cd995 : FAILURE in 6m 58s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n      ... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"a53212bd4d36873096b9334c2c6edce4de5fcb01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 19:25:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab88bc9dd8974c73b335e2202e5591a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edb54be90eba4b4687b7db037c2c05f4 : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8278227b0e045809b2ce51249242cbb : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3bc82c67d054e4cbe03759d44c8601a : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d66740ac8c0344d2bf93ef8d600f38dc : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83ac1b49d4ddea8c113e4e2f7fe7796363e9343a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 20:11:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d13fb2f230644b3bcc95686220eb7c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bae3b8257e41d3b0bf3e3cfcd782b4 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/381c9ecdb1a5492abf87f7b8e2f66cb9 : SUCCESS in 10m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4500c0c11a34c05ac7ebdfa7052ae89 : SUCCESS in 11m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3958da92762a4a99850f2b2808bbadec : SUCCESS in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0015d42c8e364e589dab402653851c30 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61968ed9ddf048429680e93634eb164c : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/7d2963cd142649629fd0a1492521f542 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7552dd8e9cc41fe9205060994f12a27 : SUCCESS in 1h 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/defb846b5cb64ea383b82cbb5c22464a : SUCCESS in 52m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa62e480f2e444682760b18e63b2a23 : SUCCESS in 18m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1794631891e2435db1d4dae12c139cd3 : SUCCESS in 17m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d9df4f624144ff495ddf2cb513c1f2b : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3da836bcc9f4299ac71528cc22a7ba9 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b50be347081e4fb591316f8f2ac327f6 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f7f908781f66482c865a75fb2ec758d0 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/46eced44e05046bf94ea84dbb1d71783 : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f111ccf151f8407d83ff5654d5872cae : SUCCESS in 7m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/993b35efb937410ca0a84cafc190ed8b : SUCCESS in 1h 54m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0557de6104314d929078a20090102d66 : SUCCESS in 1h 59m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/31c85faf3e9544d2818646e0268c608a : SUCCESS in 1h 24m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/695a3af63c3849fab3cec6919d39088b : SUCCESS in 1h 34m 28s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/279bef08fd684b9685a3f70e18cbf639 : SUCCESS in 1h 13m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a097557dbf34a3480530b86631bbe55 : SUCCESS in 1h 07m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9748a227184948d4aec318ef63e450ff : SUCCESS in 30m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a566a0f564ca4810b6b157bb31dabce0 : SUCCESS in 1h 04m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/38f98db0da1540b881fb3ba57ff56f8b : SUCCESS in 1h 23m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d853402942242b0981def0ba1953d84 : SUCCESS in 1h 16m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec8a212dd61a4246bd6cb93a538b3611 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5f5c2fdcec4407beb85a4bd9be52ada2dae06983","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 20:23:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/926733/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/85e7db2/ : RETRY_LIMIT in 4m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/926733/1/upstream-check/cinder-tempest-plugin-ontap-nfs/c876e3f/ : RETRY_LIMIT in 4m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"68e2f45317c212af70e1307c4b24fa5a97ea1709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 15:24:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91412bb6902620685192018d3ae2622a76c0dd62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 19:48:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926732 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"e1cc1d5866d13d04f9e353a5e68de6bd39751937","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-14 14:10:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f12743683baa8cb9334761f1efaed438a5d121d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 15:20:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494232aa704c4b718dcd95adc6f0413f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f49024360c824c50b49154d12ae722eb : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/818c278dd87946f49612959f3b8c3cf3 : FAILURE in 26m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7d5780994eec4858bfe6bc8b34ecafd1 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be6c1c14d4484127b4f19c2f14e541e2 : SUCCESS in 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67fa36edc4d72c5065558b91c2653fcf62c5e812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 16:23:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d2ae17d9004f58a18edeea17c8754f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efba9230436a477897c3f4f8e36b073a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb3e9e00eafd4600a3ed2020ec4f9141 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ad867330dad44328a4d9a74d140bc5c : SUCCESS in 10m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b91378db8cc146e2a556a8f0d97dfde4 : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b8f198d1b1344b29325a6a151abeb23 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b641863b45f04b1a9bb57bec662ad87b : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/955cd8f4810e4cb29a19e9cbee0e460c : SUCCESS in 57m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe5c72ef3b144286ada3f34439c63e08 : SUCCESS in 1h 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fdd7e31a7ea4e41bf83b2ac7bc3ffe1 : SUCCESS in 18m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e1fcc723f8402f81c204637b04a258 : SUCCESS in 20m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fcfdc1e0d72b4cad9089f6fec4352eb0 : SUCCESS in 16m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/faea1e05455343f0b1a3779e0a99c3c3 : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30f8a390b5644dfcb8a250ca221b201f : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13a60bdd1eeb40dfb92326c3e95d393a : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5739d3e901434bd58022b5f52497a5ab : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b230c4efa4654fed80cb40bdcde0b658 : SUCCESS in 43m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4110eac9ad2143839b33ee088ef38aad : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a45af01dabf4bd68c9812816f650675 : SUCCESS in 1h 21m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a010393a6dd0489fa53fcb1cf4441670 : FAILURE in 1h 48m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d10b2f38ca7e4df9b62176ded7736838 : SUCCESS in 1h 36m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/47fa5161585b44b0bcc3c5038d2caaa4 : SUCCESS in 2h 05m 13s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b2e77767e6a41098c7f23bbb862fdf8 : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e07a3875bf5a45918983bea4fec17ea1 : SUCCESS in 1h 08m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f298ab9ee10f4c269936aa324ed80d0a : SUCCESS in 43m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/148e329967ee4e3d91c50bba96f5defb : SUCCESS in 1h 20m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/313a49cb960a45e8b750b86aedce3352 : SUCCESS in 1h 26m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78b07b869de7418f841228e40ea568cd : SUCCESS in 1h 24m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9efe2c278604d91804fc60265693a48 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a917a08c5560196a0f92efdd9129ccf8eb402fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 16:24:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7807d58d6ab5cf837bcace0879758dc469ffdca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-14 16:29:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_09_2024_16_24_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"b12ad12e7c57b029012eea4c29f9d2b911a3f7ac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-14 17:11:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/926733/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/c124dcc/ : RETRY_LIMIT in 13m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/926733/1/upstream-check/cinder-tempest-plugin-ontap-nfs/5b8a4a7/ : SUCCESS in 45m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d639ac8cdfd835a5a7d4fd79881a911fc2e2728e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 18:05:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0f68fe3751347a4a64694f6c9d2e0b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ffacd1d959451d9bd58346a8354ef5 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b6e0986089340c6907327a12d640b2b : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96edb54d563348ad8b520b31b9819b26 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a48af6a96f964038b39de98119c04a7e : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066d19ee8e9f481d90ad945f52906db3 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/72f998e560654a33a9f6718de94308b6 : SUCCESS in 51m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f44a33e22894a13bbb993330f443be5 : SUCCESS in 1h 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71caeaab870d47d895ee55637aa1d38d : SUCCESS in 20m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09ca3ebbafce4927924b892f6ca7cc5f : SUCCESS in 14m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99f3492acc434f6d8a7f06f5df609303 : SUCCESS in 1h 02m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3c4f817b8d84a38a2ddd76de4ea6d42 : SUCCESS in 1h 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a273be2a6c84ef58cbb511dc0533e1c : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f5a6784a4600e535b0d280cff1d1b5164ae19a7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 18:05:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a438bfdbcfe7f6b72010de436df4674ce1b70a52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 18:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd76e2e49134770bf234bc62d7c5cc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b0ad74134a405e8ae8cceb3bbb7ce9 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/097e1ba3a1494d01857868e0b4a14043 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5b0c6daf6b87be3875032d5690f414423362a63","revisions":{"b5b0c6daf6b87be3875032d5690f414423362a63":{"kind":"REWORK","_number":1,"created":"2024-08-21 11:36:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/33/926733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/926733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/926733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/926733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/926733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/926733/1"}}},"commit":{"parents":[{"commit":"070ca24931e36154700119199dbf5f1606302981","subject":"Hitachi: Stop frequently REST API request in test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/070ca24931e36154700119199dbf5f1606302981"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-07-03 06:12:11.000000000","tz":540},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-08-21 11:15:20.000000000","tz":120},"subject":"Hitachi: Fix to set correct object ID for LDEV nickname","message":"Hitachi: Fix to set correct object ID for LDEV nickname\n\nThe patch fixes to set correct object ID for LDEV nickname.\n\nCloses-Bug: #2071697\nChange-Id: I2a2cfb44207b03fbb006d29ae958896f3f3231ed\n(cherry picked from commit 1895845dcf791d37dc765503b0c3300a75b16991)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b0c6daf6b87be3875032d5690f414423362a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b0c6daf6b87be3875032d5690f414423362a63"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
