)]}'
{"id":"openstack%2Fcinder~926914","triplet_id":"openstack%2Fcinder~master~I8b6fe687750b69a6d622c434a17da9fb4d13be6a","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-11 14:56:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I8b6fe687750b69a6d622c434a17da9fb4d13be6a","subject":"WIP: API: Allow newlines in snapshot descriptions","status":"NEW","created":"2024-08-22 14:00:51.000000000","updated":"2026-02-11 14:56:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"231b55ec0b643a2183b7c21ea914226f0598d23b","_number":926914,"virtual_id_number":926914,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-02 18:52:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-1,"date":"2026-02-02 21:05:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 14:01:03.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-22 14:26:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-22 15:07:47.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-22 15:31:40.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-22 15:59:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-22 16:03:47.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-22 16:10:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-22 16:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 22:45:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-23 02:31:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-26 20:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-08-26 20:59:12.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-26 22:11:14.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-27 20:49:55.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-08-28 03:33:54.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-08-28 13:05:49.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-10-07 09:36:24.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-12-02 09:17:13.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 14:00:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 16:30:07.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2026-01-21 21:48:53.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-21 22:27:50.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-21 23:24:27.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-22 01:06:45.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-22 19:45:05.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-02 16:34:48.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-02-10 07:28:39.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"}],"messages":[{"id":"1ad346048ffea8d7e57a40000caeee866658bde8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-22 14:00:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6392ba31f86e68d15e91495a9e0dd5e20b74f72b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 14:01:03.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/373/","accounts_in_message":[],"_revision_number":1},{"id":"9e09325a38f23af6ef598d8228a4a1761167f7bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 14:04:19.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"8450e18e85fffb7c5eb35413702cabf5f49ddc75","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-22 14:26:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0dd7acf3ce2251a347439113886ee08423926606","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 15:07:47.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/379/.","accounts_in_message":[],"_revision_number":1},{"id":"4fa875bd95fe1a448ded704586e72ed32f4a32c2","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-22 15:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/926914 : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c4fd2367582b1ee3fef23b20a3bb1a6c146df54d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 15:59:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a26adbff35004709bf25052a343a037a : SUCCESS in 1h 22m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6c786fbb3f224aa58366fdc77be7aa07 : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6e02d2c387d3ae2b406d458de5c0175f7ba6068f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-22 16:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"001c0a6b87272bb31d934cda4140521844a3691d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-22 16:10:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4aa83adfecba4c30bc2d9caeb89e0f74\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7feae11867542fabc31cd82a6b79212 : SUCCESS in 40m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a375f3a3de0e468880eeee30515a6fe0 : SUCCESS in 37m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/638ba05e66854222a2a6b35ce93ff5da : FAILURE in 2h 08m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed64660cef7540aea8aabf04262420c9 : SUCCESS in 35m 44s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"ff20d579c0d1d2b344e60d182d80d13c4525d5d7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 16:12:01.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/967/.","accounts_in_message":[],"_revision_number":1},{"id":"93ad67a1407e92dd90a708b0ceb39671801134f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 16:37:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5effb6a661b54145b9f047f1f8fc0229\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc49910b7714b41bded8069bd3b4c7e : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d649790ed6da4e498b214ea151e3ec0d : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c1cdb5b2b2046c9bbf573c42d6d0bb6 : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca1be50f1b0d425981b541d2edd336bc : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d2e677c1624aea81361e137de1acb3 : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/059e8bf941214e62ae0ae4607f71d11b : SUCCESS in 1h 06m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f28d4c57ee224bf5bb32d888b04700f9 : SUCCESS in 1h 31m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28f9cc1dff1b46d69047d7f8001b4a20 : SUCCESS in 55m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/06531c1f38bc4a079793bf3034c60cfb : SUCCESS in 17m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1489228f6804455b9ae1415596450f49 : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aad39f1df4bb4caab1b4a5c0fcce1005 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e80306442bc4b089814deb295ae8750 : SUCCESS in 5m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9bd9ceb9c6ea4514be71b2a063ce20a0 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a341e96564c54862bf7376fdbcb34bb0 : FAILURE in 41m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6a6c39cbbeb4618b2f398bc8fefc192 : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bfd1a72e7e644dd983ee7deef77ec10e : FAILURE in 28m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3963ec79fa104837b02fde6e9ec2a763 : FAILURE in 2h 31m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0170dfa37c2f409191286f63e5fff13c : SUCCESS in 2h 01m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bb2d5157382d4f049b114ef5c003279f : SUCCESS in 2h 18m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/09dc9dabf0734d8ab79d228ffdc8717a : SUCCESS in 40m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2043a088c3824a1da913f75c907afa59 : SUCCESS in 1h 13m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/273b5b8d3a9348bd8cf36e9e1292e2e0 : SUCCESS in 1h 29m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c3bad8fb39344d2eb05fa95d43178b19 : SUCCESS in 48m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/750f4b5f7521481cbaa29834de298342 : SUCCESS in 1h 26m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c3874b813884030a52033aedf864cd7 : SUCCESS in 1h 38m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df261a7a3786419ab354a5a7ef3ede47 : SUCCESS in 1h 46m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0716318c2c0c456abe233183d3a21f78 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1188bc03b3d19d5381534dfb0d91ab878d1c027a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 22:45:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_3.6/151e942/DellEMC_PowerFlex_3.6/1946 : SUCCESS in 2h 18m 16s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_4.x/24a6a9d/DellEMC_PowerFlex_4.x/826 : SUCCESS in 2h 00m 59s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/627aafb/DellEMC_PowerFlex_4.x_NVMeTCP/801 : FAILURE in 2h 51m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd80811cd61bb80d88c4655d15250a62402d7251","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:36:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4cdbd35dd4549f19bfbcb42812dc359\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2e40c3bfad346e98a8cf3f27010c136 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50c72d75d81c44fc87683e67e8905995 : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7eb00b8bb762bc85a986426a9773c243c7af9fc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-23 02:31:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_Unity_FC/6f21b26/DellEMC_Unity_FC/2480 : SUCCESS in 1h 55m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_Unity_iSCSI/6c96bdd/DellEMC_Unity_iSCSI/2484 : SUCCESS in 2h 03m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2bf1c3578a26218e572b61a70d8f1a05366c689","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-26 16:42:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6226550845de011a4181c189ee53cb49ea0e802","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-26 18:10:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f0c6263c06924e0ba03a5f1b533c4b6a : FAILURE in 1h 25m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a2053376788e43a29206b1553832a9a1 : FAILURE in 1h 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"449cc5ca201856cd07b92eaed0b88d13e9862ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 19:55:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cef7af913a748b6a5f60ea49b591efa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b602e72a8b948d2b8d38da12efd8e51 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c3032a05f54c79bac626556504f682 : SUCCESS in 10m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f08f5ae6ab514a2b9471ba60ef7618c7 : SUCCESS in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51b9c5edb45f47bc86a2ff51ef1429b1 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe9e24d5b3846a2adcf74b1224890c1 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/60da5fb761c44ab0b81a33454635286c : SUCCESS in 1h 04m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9704d8b720b49e089c416a55f9f1be4 : SUCCESS in 1h 26m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b077c4d10b64f83ab79ff1fc1dae69a : SUCCESS in 1h 00m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f3d734749cb49899ffd6cbaa9baed61 : SUCCESS in 19m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1089b54fd7db43c18dc4869245f0c715 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff71be14a7f84bb2ae0a7b8306dda0da : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f6dabf265ba487aa270a1a0076cdd55 : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ce6a84a07d7476a86eb158fa3b9de12 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c0d1a68275344e61a388400c32d0995a : FAILURE in 30m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0330f8bb7a244fae8a5a79a171298a59 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16e577fc5b3f4ea79b324e6f83217380 : SUCCESS in 1h 47m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f4d0b967fc0648f39d8d2b535b8f8518 : FAILURE in 2h 31m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5c08ad6f94134a12b4d51d516ac8b822 : SUCCESS in 1h 41m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a4b30aca14e0495d89eff9b2b0c5d152 : SUCCESS in 1h 36m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/919b24187f484d1291e78edc1eab58f5 : SUCCESS in 23m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ca137423ebe470781b27dfbdea746d1 : SUCCESS in 1h 11m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5e314fd121254883a425453af3ed1add : SUCCESS in 1h 34m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/166a43f5bfaf491eacb8c7690a8d5a94 : SUCCESS in 42m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab431a256b9a44539576315a585e51cd : SUCCESS in 1h 36m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7ffb3e60466f4eaf93c7d514e20d7ae3 : SUCCESS in 1h 24m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcbfe05133894eba9fc7a83827ed8a47 : SUCCESS in 1h 23m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1f5bb65c14844ecaf109a7f7ee6d1b7 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"74d23a3ecb2f0bdcc21ed53b494d68813d32fa3a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-26 20:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/14/926914/1/fc : SUCCESS in 0h 45m 09s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/14/926914/1/iscsi : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"206d24392f9398090d43ce4a0aebb10dd0f5a2f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 20:26:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa82ab8027fc45059642260c0cec1321\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e1dcebf018d4d76bfc8a17fc95c8f88 : SUCCESS in 16m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c91caeb08a24c3ebd029bfb9d25904b : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5053d361a05c835f64d7ee74d671f337ea9fec40","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-26 20:59:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/926914/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/ca35722/ : RETRY_LIMIT in 4m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/926914/1/upstream-check/cinder-tempest-plugin-ontap-nfs/19142b8/ : RETRY_LIMIT in 3m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ddb5e1582f4f3bbd480ac4d9b300ac0d656d87c8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-26 22:11:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerStore_iSCSI/0992938/DellEMC_PowerStore_iSCSI/1651 : SUCCESS in 1h 22m 28s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerStore_FC/b3ae1b5/DellEMC_PowerStore_FC/1847 : SUCCESS in 1h 04m 53s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerStore_NFS/b41e5d2/DellEMC_PowerStore_NFS/1344 : SUCCESS in 1h 21m 07s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerStore_NVMeTCP/d18df4b/DellEMC_PowerStore_NVMeTCP/1204 : SUCCESS in 2h 15m 23s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerStore_NVMeFC/49e6f30/DellEMC_PowerStore_NVMeFC/168 : SUCCESS in 1h 20m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"ffad9785796fe18f8642366ee60e0738dfc0515f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-26 22:24:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_Unity_FC/40fe0ae/DellEMC_Unity_FC/2505 : FAILURE in 2h 27m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_Unity_iSCSI/b6f0cff/DellEMC_Unity_iSCSI/2509 : SUCCESS in 2h 15m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d63b128d5e418c865b4c00f813c18c87db2ca211","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-27 12:55:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_3.6/acea804/DellEMC_PowerFlex_3.6/1968 : SUCCESS in 2h 22m 53s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_4.x/3a606c4/DellEMC_PowerFlex_4.x/848 : SUCCESS in 2h 00m 10s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/14/926914/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/e9375f2/DellEMC_PowerFlex_4.x_NVMeTCP/827 : FAILURE in 5h 53m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"002e02967a78a5083c5b0e951030214957d6aa6b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-08-27 20:49:55.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"9f2a9b4f38d08fff9d23598878d27953e442b1fc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-08-28 03:33:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/926914-1/HPMSA-iSCSI/ : SUCCESS in 3:18:57\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/926914-1/HPMSA-FC/ : FAILURE in 42:42.79","accounts_in_message":[],"_revision_number":1},{"id":"87a8c33daf78747776615ab08e5d7655b84adc14","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-08-28 13:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/926914-1/Seagate-iSCSI/ : SUCCESS in 2:21:42\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/926914-1/Seagate-FC/ : FAILURE in 2:35:09","accounts_in_message":[],"_revision_number":1},{"id":"860546163d63eb9a167466cd6e75d1dfc8a725da","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-09-04 14:07:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da52465d64d13e6098b7a5390dde5274dc6c8763","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-09-04 14:22:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b3057f91202d2fc6dba6ab042d698841b651030","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-09-04 14:38:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"508570fb90b3e4a44291efcd6e16e1fc50820b13","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-07 09:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/1/20241007143038/NEC-V-FC : SUCCESS in 4h 03m 09s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/1/20241007143038/NEC-V-ISCSI : SUCCESS in 4h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"43f0b4ca758ded32334fdcd8ec67acf1af519a00","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-02 09:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/1/20241202095414/NEC-FC : SUCCESS in 5h 50m 00s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/1/20241202095414/NEC-ISCSI : SUCCESS in 8h 22m 56s","accounts_in_message":[],"_revision_number":1},{"id":"19e25f0401417f2072b77b8ca902926f4a9449bc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-30 21:43:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f77e18a12099e712eafb95a421da9a8a474c4e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-11 13:35:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ab575acc61a8eed6a84f929de69bbd014bf857a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 13:36:01.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/554/","accounts_in_message":[],"_revision_number":2},{"id":"1352dc80cd85a651d174d6f0b0cf55ffc47da9ea","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 13:37:19.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"cef4f6e4ec2be38da18ca73152ff779f5db7ef58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-11 13:54:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23b1f689fcc90000c33fd17770a860b8501b8d3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 13:54:32.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"ddc33107b6f6e9d391c3f232c40764ac6c4cc5a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 13:54:33.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/555/","accounts_in_message":[],"_revision_number":3},{"id":"6ac04a73460a0c1a6ed289fcf83149cebd95e700","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 13:55:41.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"10c6fd18fb863e95ac5796e723f693bcde56fadc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-11 13:58:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e347dde0a06622c8814d96bc70f87543c06828f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 14:00:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I8b6fe687750b69a6d622c434a17da9fb4d13be6a .","accounts_in_message":[],"_revision_number":3},{"id":"5a394862ddb50fee62f508b8a6c071ab308b5e68","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 14:02:01.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1898/.","accounts_in_message":[],"_revision_number":2},{"id":"c9c3c1ac3ab22190c001fe191818b40e94982627","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 14:13:26.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"c94e944362051f22ad0b6025806dc4f59ab3efad","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-11 15:14:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : FAILURE in 1h 16m 17s","accounts_in_message":[],"_revision_number":3},{"id":"edde6c8903005955e4961c6c3a6b93e81ddf8b37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 15:28:37.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1764/.","accounts_in_message":[],"_revision_number":2},{"id":"916e4b6d789304f23a522ad29d9054e005816cef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:48:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8c7b03a8f9494d9784ef8e2ba5c901\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a42f88da195443d7821cf664ca6b49c7 : FAILURE in 11m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/168176be400a4cb185763c50a30ee011 : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0abc234963494f698303d76dc7896f3f : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43d420d27fa9a6bfb40cea2542aaf377b8f6f0ea","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 16:00:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d34fa8592d844272a520639000088ba4 : SUCCESS in 58m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1716a1b5cc054454b08b8deeaae049b6 : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":3},{"id":"2490cecd82b92376d4684329afdbd1ae1cb55cc9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 16:15:35.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1900/.","accounts_in_message":[],"_revision_number":3},{"id":"bdba071c0b1da7e9e1585e44e9aebf4cb8340db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 16:29:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246db22d9a874826991bc2cb7e5d0743\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/605a18bbe69e40efb97334a7029fa0f9 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ef70296c79b44a489c7759cb8e184d9 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8625df3ba0b48a7a5e435ff9d1469d1 : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/658890c5bb1f4f1998a7defecadc3c19 : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea4230fa0814814b0b915645960b4c5 : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9bcf7e91dc38469ba032e6fc5bc75fbf : SUCCESS in 52m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f784a0231d5040b8b5507e24c31b48c4 : SUCCESS in 43m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aaeca62c9c2e447ba790d01d2a03a5cf : SUCCESS in 1h 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af0e2b51ea854d93bbcac10cf2a669b2 : SUCCESS in 58m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98b47585d7b243979a3e0c9dee5e9953 : SUCCESS in 29m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2140d6a1e8648d2b2bbb5de84c49aa1 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/781d1ebe6d1f4009931f2ec98defa9fd : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f1d3404d1714f6385cdf8168e55548f : SUCCESS in 7m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a3cb264a912b45e28c0e89a2b4b2f437 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0286a53cf5f4e6eb36245cd159ab651 : FAILURE in 23m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ada48fa735e64ab0a941a42226387079 : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17200ebc8699443385d3f403472ec82a : SUCCESS in 2h 06m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/993f535e888944a49f60ee167c1cfd8a : FAILURE in 2h 30m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/487716f6feac4941a76cb154eca6e6dc : SUCCESS in 1h 51m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8b8ea318397b4484be0367bcccd170a5 : FAILURE in 18m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/85158de16f814883aa932d9290a1b375 : SUCCESS in 31m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7ea93304afe4f168f6044a9ba49e1f5 : SUCCESS in 1h 07m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8fd140eb16334236b2a2163a84edae7d : SUCCESS in 1h 26m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bad1a705f284a5d874464abad5dbafb : SUCCESS in 38m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8aea152a2d6542b3b976ac467f65b97e : SUCCESS in 1h 11m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c0d227f7fd504b1985c875407d00c018 : FAILURE in 16m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5ae4bb3b28f447091bd86850bf61936 : SUCCESS in 1h 29m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e61409df3c34a1c955ba4ad03f97e41 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"174a0e214190fe337215c3024315c65a7ba89cf4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-09-11 16:30:07.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"c7115e73b9b9101fc2c1acd339c28ed567c76faa","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-09-11 16:30:46.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"8a03f386038ef6e07e52f41fbcbde4d3a4371963","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-09-11 16:31:21.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"423fd12d36e9bd90c24b3fbc51fad4eb0bc02c62","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-09-11 16:31:58.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"bbd13ebca57f8450b06693fc3b8128d669ef3fd5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-11 17:02:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/926914 : TIMED_OUT in 3h 01m 12s","accounts_in_message":[],"_revision_number":3},{"id":"61466e01e29b297cb2f00d464e97c4be89cbd570","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 17:35:09.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1766/.","accounts_in_message":[],"_revision_number":3},{"id":"56fdf8397cc97407f47f8803ba8425eca5d3bca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-21 21:48:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"40faddd024b8b3e509c8c3caedce114f6997e017","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-21 21:48:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/29ee331120524513970e80805531bdc3 : NODE_FAILURE Node request 200-0000623307 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"c28ffb6e60275416a615c0e56b3d633eaf3569dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 22:12:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedebfedf0934d43b36076148b1f6bda\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93ca752d76564fd5a693038ad458ee0c : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e627396713aa44e3b3c623566929d328 : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4c504a934e84f39b8643495472c4c2d : RETRY_LIMIT in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30d1387667fc915bcdc810a891bb4c1e7d4ab5d0","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-21 22:27:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1ccb7e5d35724f6ebb101f73a41129bb : FAILURE in 38m 48s","accounts_in_message":[],"_revision_number":4},{"id":"e141249ab79e1c2806a1cd9ba92215263ad7afeb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-21 22:44:47.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8b6fe687750b69a6d622c434a17da9fb4d13be6a .","accounts_in_message":[],"_revision_number":4},{"id":"d8d07494002b70d042c3bd9e6fcc4a9d21d68b3b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-21 23:09:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : FAILURE in 1h 19m 54s","accounts_in_message":[],"_revision_number":4},{"id":"c747622b30cfcf7576337657261a71d82c49ed16","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-21 23:13:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b6ac833c51df45ee935f35c3e308d578 : FAILURE in 1h 07m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d9d6f1b75b5c431eb53035f99517baab : RETRY_LIMIT in 15m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5c7532d38063b623c2703b1df6d5c3e8dcbfdcb3","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-21 23:24:27.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_320/index.html : SUCCESS in 95m","accounts_in_message":[],"_revision_number":4},{"id":"03218a4a937444ea8548943ba2884ebdcc643a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 00:59:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e7102d93bd44e5a91d129c24d87014e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fccb78465df44c908f3a828b793b1086 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34879328b46c45d7959e8f38dc7d301d : SUCCESS in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37cabfef35684c78a607f7d1c899b4d8 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8212f8c9a6645f487a6ddef5eee9380 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c521609badda4a5188217eb9b09dc5ec : SUCCESS in 29m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f01c59dcbab422184d80bb408d76c9c : SUCCESS in 35m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9443cef16d70451ba66b6d40f593a5a8 : SUCCESS in 1h 23m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61a7f6def6f94bb7b03b5270beb204ca : SUCCESS in 47m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab2e58585f7d4d0990c77d76c1d42ee5 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c32345bd5a004afdbe260a7aa6b4aaca : SUCCESS in 3m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1918116a729a49148b9f96fa4b09ba2a : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02e2da36481c4fce90f103e1f442e8d4 : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/afbac75fd62f4f3f9e3c6dfaf82adfb5 : SUCCESS in 4m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc6d03b6eac342a3a741a983b5ee7844 : FAILURE in 18m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f9e0142e6fb4f06b1e33e07c13f5194 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da8679b21b3f4b7eb2f23ac35805d402 : SUCCESS in 1h 54m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/059a4f91116c45d59539550f50b7e493 : FAILURE in 2h 20m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/326591532424479f9ad0e76dc16cdfc1 : SUCCESS in 1h 44m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/12194ba91f534a83831cd818910996f9 : TIMED_OUT in 3h 06m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dd3569f7b7c54e0daa817a75ac8e17e1 : SUCCESS in 37m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8b19d1ce7a274176bee3b1292914f6b5 : SUCCESS in 59m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b623590a2e1247c19645d1581a860271 : SUCCESS in 1h 24m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43f1a8e1dc7944fbae1da67293cec0fc : SUCCESS in 49m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b12f57a81f6f4b9e819524f39f24123b : SUCCESS in 1h 46m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9b24c49cc985473eac8238532dfc9bd8 : SUCCESS in 1h 24m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1352ac491ae94731bc3d3002bced5e04 : SUCCESS in 55m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9d1f290a243406bbfe7287948c23a01 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":4},{"id":"a39f46f731d4e6b5406f8d7af9e979461972b160","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-22 01:06:45.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_01_2026_00_59_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"5c503e58955291563d6d883c3eda64fbe0f18674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-22 14:01:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"427115cdec708d16e6f7156b68b476bbb0081b17","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 14:01:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/f8e98843716447ea8ea6f059e5341e48 : NODE_FAILURE Node request 200-0000623322 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"340ed06fdd1cc2ced8c91e2d817d3239c2180c5a","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-22 14:44:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1ae912a131da451f9b4d243aaa01b630 : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":5},{"id":"e0fe98739de4223ab94b5e5dbc4aee19805d0fdb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 15:04:18.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8b6fe687750b69a6d622c434a17da9fb4d13be6a .","accounts_in_message":[],"_revision_number":5},{"id":"eb6976fc46fcd0a4ecb568afe115eb0fe42e1f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 17:02:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea4e29541e61427ba81e1b00daea13f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1580f4edf3ac45f98bd739bff85f9e67 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70b821be6b744040859b1568d71c12f7 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7d0eafd43fd402ca158e02fdf0b6c9c : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a386fa3e3474bf9ac78afaad4d11b25 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5aa6ada31ed341b59dd8701e2406f236 : SUCCESS in 1h 05m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/078646abe81c4971b9bbdb8eadf576c3 : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44c1d2775aff4ef28280ca63f0eef60e : SUCCESS in 1h 22m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4db8be4424e4f52a6bb9e159e54efb8 : SUCCESS in 59m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7052ddfecc9479ea37372e467ae2a76 : SUCCESS in 28m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d3ad7c429ce405db03da5f50146603e : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/502a4d86af434141b38a1cdbda6bb27f : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a602b07673a4185b8ea83d12beb34a1 : FAILURE in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f4ad69be793b41e2a1f0e85faa0fe893 : FAILURE in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/47d4e21a8ede4ad09b4d392e81040dc0 : SUCCESS in 43m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8df7efcac0364b7e9e63f7b240f6ddd2 : SUCCESS in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22ff763e7d7f4ab6b511dc28bdaed765 : SUCCESS in 1h 43m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d1a63494387a49e1bc9cbd76e31facd8 : FAILURE in 2h 34m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64cc2e9d220a414b83bc20b83a1200ac : SUCCESS in 1h 43m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2146fdf9289541568f7eff9c5b995425 : SUCCESS in 1h 55m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/501252807474413185d8c812d0d08978 : SUCCESS in 37m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c201d086cff84ca3ae938d3367a5e2df : SUCCESS in 1h 20m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c953fddc15c49e99a649792b98cf68f : SUCCESS in 1h 39m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b05e76b0a4f4073996dfb159a1b5ecd : SUCCESS in 44m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3138e1a26be547aabc4306e161fbd12e : SUCCESS in 1h 56m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8f913a8685a4432fb4fa338636067684 : SUCCESS in 1h 59m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f90bf6eb8aee46cebe881006bfab348a : SUCCESS in 1h 38m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0b6e31b5cb40b3bae823bdc6e5186d : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":5},{"id":"91ec2b81410a1cd6497024d16ed3a562890a80d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 18:10:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/932fec43f12848828c6cd3a88fda076c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/925f5880990047358c824f39ec36e162 : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92c5d853bae0408e80b887a7eec0df82 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4be8c5e748244d85b0afad5044a4da17 : RETRY_LIMIT in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4daacb23904480b9e10ffaf06dd5272af4377b5d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-22 18:30:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : FAILURE in 1h 18m 57s","accounts_in_message":[],"_revision_number":5},{"id":"45ddcc41e17e03cdec1c7bc610e77600a1790a61","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-22 19:45:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-14-926914-4-fc : FAILURE in 2h 7m 25s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-14-926914-4-iscsi : FAILURE in 1h 46m 11s","accounts_in_message":[],"_revision_number":4},{"id":"d0d10e702f6fbd01068bed8b1f087ef03085a1f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-22 20:30:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f633d6f1a79b8471161b51b7da460a4b4fa48a34","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 20:30:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a5f285ac961449fda8d23c93c4c8cf4f : NODE_FAILURE Node request 200-0000623323 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"2db51e2bca2f9bb4d1395d5078b92ba608c5e688","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-22 20:35:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c22c85d158089c574d23a972012b6a74413a5652","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-22 21:09:25.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/5f9413d127b84356b44ffb2f6575ef40 : FAILURE in 38m 32s","accounts_in_message":[],"_revision_number":6},{"id":"ed68742e49ef4a14b4e6f765d877d75e4611ca67","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 21:31:37.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8b6fe687750b69a6d622c434a17da9fb4d13be6a .","accounts_in_message":[],"_revision_number":6},{"id":"9592ec41fce152df915f38c033d405af4c6265e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-22 21:34:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3e7062b3cc994926975a80ec0312682b : SUCCESS in 45m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e4c82a60dc6441a9b8338c2a6df1dd2c : RETRY_LIMIT in 15m 18s","accounts_in_message":[],"_revision_number":6},{"id":"da14983010bcdaed5f6d31b55be5dc7905678be2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-22 21:50:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : FAILURE in 1h 18m 37s","accounts_in_message":[],"_revision_number":6},{"id":"89fa8a3c94ff7013d75123e7a39502d268f1b2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 23:11:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00e4ac956b194958ba74d4f511029e0d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/925ea0af8db84f3192a99936a83f522d : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23438fb4ce2b4967994e2bd933982921 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36fb090c7ccd4463a1e362c90015d82d : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7032a259532842c5ac6edcca4e2c5d8d : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371d6e5960bb4bc8bdc752a6c2bccfbc : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/244237690056415fb3432f9a69edca43 : SUCCESS in 1h 00m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5104f51d0d1b47fa9c64766e7ca4eefd : SUCCESS in 34m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0433e77c30744122a679c107c4372bc0 : SUCCESS in 1h 32m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4c7bdec94334fdf8147627da5bad9c2 : SUCCESS in 36m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f173a177ed1433cbe36fc9e28107695 : SUCCESS in 19m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5b5ef0919884ce3b8a3d3a08314b34b : SUCCESS in 6m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85c3d9991c5540ac940bc9d90c07b759 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/011276a80caf45fca813388ef2b9b143 : FAILURE in 3m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/df46410bb81a4af9817c848efdb502b3 : FAILURE in 3m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a963f508318b46e18b95babad5e720da : FAILURE in 40m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/475c0f7c537643d9bf1fec5f444bec3d : SUCCESS in 3m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/139c5397ea394d3b87c8d9507c8f82c7 : SUCCESS in 1h 40m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e23c50ff32d247ce8ffceb71261d84f6 : FAILURE in 2h 35m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd5792ca66184ecebd188318ff709719 : SUCCESS in 2h 00m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b912c2e3af7041c2b66c9df9c0dc7f80 : SUCCESS in 1h 54m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f2bb9cd10c604226857b8acc62dabf07 : SUCCESS in 29m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e23d4cb1eb774cb7bb704cedfd4c1b67 : SUCCESS in 1h 15m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0d2510b6e09f408a8785ae3297a3e3a6 : SUCCESS in 1h 31m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6425279d92a9421897849844cf3fcb7b : SUCCESS in 40m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e8aebbc01bc47a3868e85dc64a45fce : SUCCESS in 1h 11m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2251564e1f0d42688757dedae02a9469 : FAILURE in 1h 07m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8aacf59410864ee38ff175d750a6fe86 : SUCCESS in 1h 25m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38482850decc4c5aa2c47258febb0e1e : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":6},{"id":"40aa10af1c6b0621cc2cc195df0e3b230a60587b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 01:13:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805d499d064a4690adcee7cb2d8914dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1433b235f1646a19aefd387cb1a8fae : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa6eaaa5f756425a9f5cde4a4be729c7 : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/80a107d19d80442c86f1a5c017467db7 : RETRY_LIMIT in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74a8573cb526c2e2300c5432bb5791906c1e1a63","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-23 05:08:06.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_337/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":6},{"id":"2ca37305240ab45267e285dac853ba30b1735a22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-02 16:34:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"580e384514f572dd504a7015b1d46bb9e2f5725e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-02 16:34:48.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/775/","accounts_in_message":[],"_revision_number":7},{"id":"62643c9d6bc809e20a9d149614cc3a104a588d71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-02 16:34:52.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/926914/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"6588109d87a7b96ac79839c6c6d4a789a1ba4342","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-02 16:35:33.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9b6fef8c633859ca793ddeb393b7c6ce387a95ad","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-02 18:05:13.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8b6fe687750b69a6d622c434a17da9fb4d13be6a .","accounts_in_message":[],"_revision_number":7},{"id":"b7e453574f2195884c470b6dd300052f16a39fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:52:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587aa5dcebf044e781099a5da6166c61\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/588f655d5128468a9f117d914dee5ea3 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec672bfa6f26406ab804ffc83c942790 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/738c5dee49be467d83157e7d520df74f : SUCCESS in 7m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f82d2ceeb107450d9edb6ff20fd49045 : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57549e195542467b96526721d80a5b1a : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/53e3512d95e245888551cc698ca4c6b8 : SUCCESS in 30m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af6af283e5734efda9f9817505dd4865 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/446b97ed51474fd596cc3d5c33eb9db2 : SUCCESS in 1h 49m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef6b925ef4564aea96a3dca42356d93b : SUCCESS in 33m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22bcc69e1e394386913c92b7b4ae93a8 : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd9f892ea6ea40aab85330a4f5fce545 : SUCCESS in 6m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20571c9dfa9e4d0587e9ce465a62308b : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6bea6506feb4c13856bd4a513cf2c8d : SUCCESS in 5m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/27aabc7542d2489382727d634550edc5 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d9d3bd6c33164a7ba1be34e4ec8e9d05 : FAILURE in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ea763d3a19348c58fc76fa26a7f6146 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0cd285f40a54b3caa5ce71c8c17c157 : SUCCESS in 2h 11m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e724d174b9d47d0b065a780e2a11604 : FAILURE in 2h 15m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e86dbd2c49c5439f9467960ace3c4ecc : SUCCESS in 1h 42m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2b8bd8f71c3f40868237c3d3629147bd : SUCCESS in 1h 31m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e9570bfb8d744d7ba1f92a6d52b50b55 : SUCCESS in 16m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2682244d0781466798c84b9e43a2a8ef : SUCCESS in 1h 09m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2342089898b46b18037d0da0ac89c2a : SUCCESS in 1h 46m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e1c046362cb746418ac2f45fcc0cad40 : SUCCESS in 44m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e2cecca2de9147f2b897436472043e35 : SUCCESS in 1h 56m 41s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2a3ff4da1ef74e02960ea0ead31f9b87 : SUCCESS in 1h 41m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9338c0542ce54c1fa24de321abf27e79 : SUCCESS in 2h 10m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aa99db3ab634bb39d924171a4b04e97 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":7},{"id":"4a8650579be6f265414c13d114ab3d143f6839db","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-02 18:54:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6a2afb638a854defb6dd3995f79c39e8 : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":7},{"id":"855aa30a5e5457d9ec7f5ffcc65b675ee144c3ce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-02 18:55:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/926914 : FAILURE in 40m 19s","accounts_in_message":[],"_revision_number":7},{"id":"2e896eba5fea287ae869f2fbe6416a1cf4cda236","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-02 19:01:12.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_02_2026_18_54_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"e349a128e736feda00a8980045ba1af6021888bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:53:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67527db7df3d48169861cddba407b5f7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ba248f27c684d1fa1620163bdc600b2 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b28adf5bef044b5b21ce49a39055211 : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3175604bd57e4626b111b8758875800f : RETRY_LIMIT in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ef88e65f435a1c68025a003ba163acdb4d93a064","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-02 21:05:22.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6b73a92c2e296ce445b567676332c1ed082e7792","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-03 10:19:56.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_470/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":7},{"id":"3227ec7557d761c9bffeeb87feff322aa18e17c4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-09 14:57:32.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/7/20260209134044/NEC-ISCSI : FAILURE in 9h 47m 58s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/7/20260209134044/NEC-FC : FAILURE in 10h 16m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d957bb764eb95910d7971996c2eedf74e946cc88","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-02-10 07:28:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"570d04af20b917082d3fbed84eb75e8b6b113a36","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2026-02-10 18:08:32.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/7/20260210225825/NEC-V-ISCSI : FAILURE in 3h 25m 43s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/14/926914/7/20260210225825/NEC-V-FC : FAILURE in 4h 10m 04s","accounts_in_message":[],"_revision_number":7},{"id":"231b55ec0b643a2183b7c21ea914226f0598d23b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-11 14:56:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4d62185bb0ef889b4471f92b92665e2b7d4ae0ca","revisions":{"ab2200efa2fc46c6b3e77088b951149368a65bd8":{"kind":"REWORK","_number":1,"created":"2024-08-22 14:00:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/1"}}},"commit":{"parents":[{"commit":"221a66940ebbba97ecfc4f8830cea50adb3a97a3","subject":"Merge \"Hitachi: Prevent to delete a LDEV assigned to multi objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221a66940ebbba97ecfc4f8830cea50adb3a97a3"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:48:51.000000000","tz":0},"subject":"API: Allow newlines in snapshot descriptions","message":"API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab2200efa2fc46c6b3e77088b951149368a65bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab2200efa2fc46c6b3e77088b951149368a65bd8"}]},"branch":"refs/heads/master"},"122f0e492df322584ec379ca86dc15b8c7ed0afc":{"kind":"REWORK","_number":2,"created":"2025-09-11 13:35:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/2"}}},"commit":{"parents":[{"commit":"ffa5996381e4667dd1f7043b3c842edfb49b881b","subject":"Merge \"NetApp: iSCSI/FC detach operation fails when multiple initiators connected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffa5996381e4667dd1f7043b3c842edfb49b881b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-09-11 13:36:11.000000000","tz":0},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/122f0e492df322584ec379ca86dc15b8c7ed0afc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/122f0e492df322584ec379ca86dc15b8c7ed0afc"}]},"branch":"refs/heads/master"},"2cf090dc36e656f366986517c10f6112ce67e400":{"kind":"REWORK","_number":3,"created":"2025-09-11 13:54:13.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/3"}}},"commit":{"parents":[{"commit":"ffa5996381e4667dd1f7043b3c842edfb49b881b","subject":"Merge \"NetApp: iSCSI/FC detach operation fails when multiple initiators connected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ffa5996381e4667dd1f7043b3c842edfb49b881b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-09-11 13:54:13.000000000","tz":0},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cf090dc36e656f366986517c10f6112ce67e400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cf090dc36e656f366986517c10f6112ce67e400"}]},"branch":"refs/heads/master"},"0eb82aa60352bfc1820eecefe7152650f4904841":{"kind":"REWORK","_number":4,"created":"2026-01-21 21:48:28.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/4"}}},"commit":{"parents":[{"commit":"efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753","subject":"Merge \"VAST Data Volume Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-01-21 21:48:08.000000000","tz":0},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0eb82aa60352bfc1820eecefe7152650f4904841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0eb82aa60352bfc1820eecefe7152650f4904841"}]},"branch":"refs/heads/master"},"929c40d00c35c144b807d5ee8595f8f063c69da0":{"kind":"REWORK","_number":5,"created":"2026-01-22 14:01:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/5"}}},"commit":{"parents":[{"commit":"efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753","subject":"Merge \"VAST Data Volume Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-01-22 14:01:07.000000000","tz":0},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/929c40d00c35c144b807d5ee8595f8f063c69da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/929c40d00c35c144b807d5ee8595f8f063c69da0"}]},"branch":"refs/heads/master"},"427b870bc9f72b974e53af386d01597d0403779e":{"kind":"REWORK","_number":6,"created":"2026-01-22 20:30:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/6"}}},"commit":{"parents":[{"commit":"efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753","subject":"Merge \"VAST Data Volume Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-01-22 20:29:24.000000000","tz":-300},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/427b870bc9f72b974e53af386d01597d0403779e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/427b870bc9f72b974e53af386d01597d0403779e"}]},"branch":"refs/heads/master"},"4d62185bb0ef889b4471f92b92665e2b7d4ae0ca":{"kind":"REWORK","_number":7,"created":"2026-02-02 16:34:38.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/14/926914/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/926914/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/926914/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/926914/7"}}},"commit":{"parents":[{"commit":"efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753","subject":"Merge \"VAST Data Volume Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc3217d8b3ce8f4f1a66c7d5b7c940d73a36753"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-22 13:31:56.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-02 16:34:12.000000000","tz":-300},"subject":"WIP: API: Allow newlines in snapshot descriptions","message":"WIP: API: Allow newlines in snapshot descriptions\n\nVolumes allow newlines in descriptions, do the same for\nsnapshots.\n\nThis takes effect with new microversion 3.72.\n\nThis leaves the difference that volumes allow newlines\nin names and snapshots do not, but I think that case\nneeds more careful evaluation.\n\nTODO: testing, tests\n\nCloses-Bug: #2021909\nChange-Id: I8b6fe687750b69a6d622c434a17da9fb4d13be6a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d62185bb0ef889b4471f92b92665e2b7d4ae0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d62185bb0ef889b4471f92b92665e2b7d4ae0ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
