)]}'
{"id":"openstack%2Fcinder~927069","triplet_id":"openstack%2Fcinder~master~Ib103160250bb3d6f2e0a2fbb057601c31daa817a","project":"openstack/cinder","branch":"master","topic":"hitachi_url_broken","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2025-10-17 14:30:26.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-10-17 14:10:10.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}},"hashtags":[],"change_id":"Ib103160250bb3d6f2e0a2fbb057601c31daa817a","subject":"Hitachi: Correction of Broken URL Links in the User Guide","status":"MERGED","created":"2024-08-26 04:29:14.000000000","updated":"2025-10-17 14:31:51.000000000","submitted":"2025-10-17 14:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927069-hitachi_url_broken","meta_rev_id":"6bee956ced81fc55ca6a621951aa3f0b6a7d2fc9","_number":927069,"virtual_id_number":927069,"owner":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 14:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-17 14:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-10-17 14:10:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37269,"deleted":true}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 04:29:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-26 04:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 04:45:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-26 05:20:52.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-26 05:32:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-26 05:38:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-08-26 08:02:32.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-26 08:40:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-26 09:19:02.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-26 10:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-08-26 14:38:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-26 19:14:25.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-08-26 19:53:44.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-08-26 23:53:02.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-08-27 19:06:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-24 05:18:01.000000000","updated_by":{"_account_id":37269,"deleted":true},"reviewer":{"_account_id":37269,"deleted":true},"state":"CC"},{"updated":"2024-09-24 05:42:20.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-26 13:10:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"51dce471da4e68421af3d982c346beaadc03e85b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-26 04:29:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e158d317b2855a86d96114057aaec881d98aca3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-26 04:29:23.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/389/","accounts_in_message":[],"_revision_number":1},{"id":"7a9cc254149a99d9db53fa474f576c742d67bd96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-26 04:34:34.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"867a77a204f24fb99681c9d995a441936fae5aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 04:41:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33376db5c2b24fa8bb5f4256a3053bba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de913cfb4b64e8ba24dad6521ff7a0c : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be722fc2e184ad59842042e679ae916 : SUCCESS in 12m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/30d73897ccb94cbe95131bfdcb3ac61e : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6e72b586f5417eea7dcb9b9a29125227aab6e811","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-26 04:45:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5eb521864d0df1ad0f9de217e5eb36be47bc39cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-26 05:20:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/399/.","accounts_in_message":[],"_revision_number":1},{"id":"ac1f1ec927ed6bdbdebc6f148b8df7a683372c4f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-26 05:32:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/927069/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/228cc5a/ : RETRY_LIMIT in 3m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/927069/1/upstream-check/cinder-tempest-plugin-ontap-nfs/9bd81e2/ : SUCCESS in 44m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee4d62127c42eff8e8e4a59d688392ad282018da","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-26 05:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927069 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"caeeae3c39f57a0a8db34e0d5bab859642f7962d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-26 06:49:21.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/984/.","accounts_in_message":[],"_revision_number":1},{"id":"52ffdffd489491c4eaaf08bdba9e575fa2c0d595","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-26 08:02:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerFlex_3.6/7de2d15/DellEMC_PowerFlex_3.6/1961 : ABORTED in 1h 12m 37s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerFlex_4.x/edbdb7f/DellEMC_PowerFlex_4.x/841 : ABORTED in 1h 02m 14s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/9b261c4/DellEMC_PowerFlex_4.x_NVMeTCP/820 : ABORTED in 49m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"21e1780009a111ab5e5fd64db2f5c78f0d4acf5f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-26 08:40:37.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_Unity_FC/5e90247/DellEMC_Unity_FC/2500 : SUCCESS in 2h 09m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_Unity_iSCSI/997ce9e/DellEMC_Unity_iSCSI/2504 : SUCCESS in 2h 06m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"349394b8c7b13893f096ee87aac1f9591369dce0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-26 09:19:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerStore_iSCSI/3c86f38/DellEMC_PowerStore_iSCSI/1647 : SUCCESS in 1h 18m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerStore_FC/faa5b6e/DellEMC_PowerStore_FC/1843 : SUCCESS in 1h 19m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerStore_NFS/c64694d/DellEMC_PowerStore_NFS/1340 : SUCCESS in 1h 10m 27s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerStore_NVMeTCP/a34ad12/DellEMC_PowerStore_NVMeTCP/1200 : FAILURE in 2h 47m 37s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/69/927069/1/check/DellEMC_PowerStore_NVMeFC/cad4127/DellEMC_PowerStore_NVMeFC/164 : SUCCESS in 1h 14m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"99bc7342397f6c6c1c874c5a4d8700cadd444dfc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-26 10:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/1/fc : SUCCESS in 0h 47m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/1/iscsi : SUCCESS in 0h 59m 35s","accounts_in_message":[],"_revision_number":1},{"id":"51cf6fe2acdcecf650379ead1deeb6f78941e4b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-26 11:03:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/988/.","accounts_in_message":[],"_revision_number":1},{"id":"5707aac0fd3db9285f8435cc46f16e696b6566f6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-26 14:38:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/1/20240826194118/NEC-V-ISCSI : SUCCESS in 3h 39m 25s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/1/20240826194118/NEC-V-FC : SUCCESS in 3h 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0d0afa72c87d35e0f035e1a9fef1aabb47412bcb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-08-26 19:14:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/927069-1/Lenovo-iSCSI/ : SUCCESS in 2:23:30\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/927069-1/Lenovo-FC/ : SUCCESS in 2:17:23","accounts_in_message":[],"_revision_number":1},{"id":"d01377119c29e65dab9a0dfb39a0950bfa1ac515","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-08-26 19:53:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/927069-1/Seagate-iSCSI/ : SUCCESS in 2:24:46\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/927069-1/Seagate-FC/ : SUCCESS in 2:16:01","accounts_in_message":[],"_revision_number":1},{"id":"98cdbbac9c3f0d8fbd28bd169536c12af2296c74","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-08-26 23:53:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/927069-1/HPMSA-iSCSI/ : SUCCESS in 3:19:05\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/927069-1/HPMSA-FC/ : FAILURE in 42:18.43","accounts_in_message":[],"_revision_number":1},{"id":"65917118291e3668c9d545b9c7d1decdba0db189","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-27 19:06:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/1/20240827221933/NEC-ISCSI : SUCCESS in 5h 19m 42s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/1/20240827221933/NEC-FC : SUCCESS in 5h 47m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7b6a31d46af7c73bd04f4902855572018142bb97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-24 05:17:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e698ebba87e40899a68c47ee3c22a4e73571859","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-24 05:17:58.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/616/","accounts_in_message":[],"_revision_number":2},{"id":"8c92f44282d99ff7a156b1b63870e99e58f16c74","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-09-24 05:18:01.000000000","message":"Patch Set 2:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/15/","accounts_in_message":[],"_revision_number":2},{"id":"58b6716a3c1bfb05cc9dd34184038ae44a601480","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-24 05:22:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"b6dbce11153986a7898168722b6050d121b91ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 05:28:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85073e8e77254512a3c0dc86443b3ac6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e249a5b56989451c8c822f97950af2d6 : FAILURE in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb01889f4e84551b72dd9c0622ca20d : FAILURE in 5m 28s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b590290b9454495c92c8c49e343908de : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"273f4dfec4b035d55220c780c0f5f2c540fcead3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-24 05:31:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927069 : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9a42bfcea6f2688612af54b709f45fb41170f249","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-24 05:42:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/927069 : FAILURE in 23m 00s","accounts_in_message":[],"_revision_number":2},{"id":"db7e1a1a009ab847a745729400047376765840ce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-24 05:47:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927069 : FAILURE in 28m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f963a45e2a05f8bc750db19f404b7f73e1c01d07","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-24 06:10:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/583/.","accounts_in_message":[],"_revision_number":2},{"id":"0662fc4daa35061b28360d248b4423bffb87ec1c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-24 06:40:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1172/.","accounts_in_message":[],"_revision_number":2},{"id":"96f26fe36bf0a453d68f1c6fee3ac656ecfa2fe3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-24 09:00:11.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/2/fc : SUCCESS in 0h 47m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/2/iscsi : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3aa6ff7346bae8080edb1c77808a47521b2a8b75","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-24 15:25:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/927069/2/check/DellEMC_Unity_FC/309d60b/DellEMC_Unity_FC/2647 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/2/check/DellEMC_Unity_iSCSI/c299499/DellEMC_Unity_iSCSI/2656 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"44cd41ea87b7a5bb0ba1334cc826b970c53dacc4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-24 15:52:39.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/2/20240924163955/NEC-V-ISCSI : FAILURE in 7h 27m 54s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/2/20240924163955/NEC-V-FC : FAILURE in 8h 12m 41s","accounts_in_message":[],"_revision_number":2},{"id":"fa5ee8aa917baa45b88e75ec090d220476cf1017","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-24 19:59:33.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/2/20240924192248/NEC-ISCSI : FAILURE in 8h 58m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/2/20240924192248/NEC-FC : FAILURE in 9h 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"388f9307aa78e9b2b5f3e13b9ba10740690652e9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-25 22:12:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/69/927069/2/check/DellEMC_PowerFlex_3.6/fa25f22/DellEMC_PowerFlex_3.6/2140 : SUCCESS in 2h 28m 15s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/69/927069/2/check/DellEMC_PowerFlex_4.x/7e2797b/DellEMC_PowerFlex_4.x/1024 : SUCCESS in 2h 08m 30s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/b58f151/DellEMC_PowerFlex_4.x_NVMeTCP/964 : FAILURE in 5h 53m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0e2daeaea625b93da6db09959cc2108b2d32d9dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-26 06:04:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"54ba4ddb7630cc7fe0a69167f8d19b442adc55fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-26 06:04:13.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/620/","accounts_in_message":[],"_revision_number":3},{"id":"499973edc48d576954c4f0bf6da230fd7c4b21e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-09-26 06:04:16.000000000","message":"Patch Set 3:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/17/","accounts_in_message":[],"_revision_number":3},{"id":"c51b698f3e3f411548f3d9add382044ece0c9d2a","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-26 06:05:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"0b1d165106559d6deb88b3fc0090dfba59ca4f06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-26 06:09:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"3522901a6b207aff91fc3a145afce70337c7f029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 06:14:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/665dd7c64d244cd1bdaa8f4f49749c02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ed736230294396aac670afb0f6c51d : FAILURE in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7fa22cc006943c49d784b2501eff1cf : FAILURE in 4m 43s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/97e535cb9d134c26865a462f55c86640 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"39c10a3dc7180a084a9258c4e0690f19b5e3f8f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-26 06:28:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927069 : FAILURE in 22m 36s","accounts_in_message":[],"_revision_number":3},{"id":"54e7f07e56df34f7db2683449afa2f79ffc8c636","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-26 06:40:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927069 : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":3},{"id":"79b8245d85e7d4e95f3674d3bb021aeb378f4dda","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-26 10:00:08.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/3/fc : SUCCESS in 0h 46m 49s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/3/iscsi : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":3},{"id":"fd1ee32825fb9180df02990a218920cc90b22c09","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-26 12:04:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/927069/3/check/DellEMC_Unity_FC/af00d9d/DellEMC_Unity_FC/2651 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/3/check/DellEMC_Unity_iSCSI/163f034/DellEMC_Unity_iSCSI/2660 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e135cff41dd79bec1a1908a9dfb6368690e68aad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-26 13:10:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e8de04f8d18c4956e929755a469ff7f723be4e9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-26 17:13:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/3/20240926172128/NEC-ISCSI : SUCCESS in 5h 14m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/3/20240926172128/NEC-FC : SUCCESS in 8h 52m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f7535177e7644a16b07ef7ccfa673bf191000c0e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-26 18:09:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/3/20240926224045/NEC-V-FC : SUCCESS in 4h 19m 41s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/3/20240926224045/NEC-V-ISCSI : SUCCESS in 4h 28m 55s","accounts_in_message":[],"_revision_number":3},{"id":"56d88718fb10d3898b1e10334c185faffa1ba0c1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-26 21:49:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/69/927069/3/check/DellEMC_PowerFlex_3.6/0b7394c/DellEMC_PowerFlex_3.6/2144 : SUCCESS in 2h 17m 04s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/69/927069/3/check/DellEMC_PowerFlex_4.x/75dc2b5/DellEMC_PowerFlex_4.x/1028 : SUCCESS in 2h 13m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/57c34ee/DellEMC_PowerFlex_4.x_NVMeTCP/968 : FAILURE in 5h 54m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"cd662ef80fb35dafbabb356256ce5cfa7f292157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-27 08:41:12.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c412e043d4f4610a8b8ab6d9757c26298a1d30c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-27 08:41:22.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/634/","accounts_in_message":[],"_revision_number":4},{"id":"f371c3782c1369e4f12546a20176c99689e2bb23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-09-27 08:41:25.000000000","message":"Patch Set 4:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/29/","accounts_in_message":[],"_revision_number":4},{"id":"049561d30db543d6a17a280a6fcc1a400ce02f42","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-27 08:42:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"5cb5bd5b4a8b75a9a3c50cbdf9e66b62f82431ae","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-27 08:42:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"01fd65ecad50faf43cb63c33cb3d9af140501abd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-27 08:47:23.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927069/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"a725e33d75b24dd8052011448620b9273436f119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 09:01:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd07c00e04c34c5eab85d088f08c1c27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d24c31070044540a2fc5cc8144ef0d4 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726835e52d354a2b9c52fcffe6d1d899 : SUCCESS in 12m 10s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04c17ace3d0c4f29a190fe506c589876 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e0182372ea3cbcb0bff507247b0e5c6d22cf07c9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-27 09:17:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/927069/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/7bee833/ : RETRY_LIMIT in 8m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/927069/4/upstream-check/cinder-tempest-plugin-ontap-nfs/4c38144/ : RETRY_LIMIT in 8m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"6320f91b15d38fd1609df48970594400a01c0a17","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-27 11:09:17.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerStore_iSCSI/b538da7/DellEMC_PowerStore_iSCSI/1767 : SUCCESS in 1h 22m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerStore_FC/4da0fb7/DellEMC_PowerStore_FC/1965 : SUCCESS in 1h 28m 47s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerStore_NFS/f5fee71/DellEMC_PowerStore_NFS/1460 : SUCCESS in 1h 21m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerStore_NVMeTCP/fbc652a/DellEMC_PowerStore_NVMeTCP/1319 : SUCCESS in 2h 07m 06s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerStore_NVMeFC/2a48999/DellEMC_PowerStore_NVMeFC/284 : SUCCESS in 1h 23m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"84060e569ac50ff9ee5778e91dbded89e563f4c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-27 11:24:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927069 : FAILURE in 43m 31s","accounts_in_message":[],"_revision_number":4},{"id":"5ace8db2c3b55d08a7cbc88967f26cb8a728f2d8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-27 12:28:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927069 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":4},{"id":"02fa9ee6029ca4df2a1944bd39547bdfb4d52351","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-27 13:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/4/fc : SUCCESS in 0h 53m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/927069/4/iscsi : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9d80c33051c8445d1c9efd4ecad28ece9037df0d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-27 19:43:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/4/20240928002933/NEC-V-ISCSI : SUCCESS in 4h 12m 25s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/4/20240928002933/NEC-V-FC : SUCCESS in 4h 13m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ff6ec33c65baeccfd7ba388387207b550f8f6c32","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-28 15:07:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_Unity_FC/4b5cfb9/DellEMC_Unity_FC/2659 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_Unity_iSCSI/c744bd9/DellEMC_Unity_iSCSI/2668 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"135036f86664529d925a3aa9a1ac400a917439b4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-28 16:31:41.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerFlex_3.6/68b907b/DellEMC_PowerFlex_3.6/2152 : SUCCESS in 2h 04m 15s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerFlex_4.x/d423c5b/DellEMC_PowerFlex_4.x/1036 : SUCCESS in 2h 01m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/69/927069/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/ec4f40e/DellEMC_PowerFlex_4.x_NVMeTCP/976 : FAILURE in 5h 53m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"0fbf919751afa17810807abd46ef8258c453e845","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-29 04:05:33.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/4/20240929035217/NEC-ISCSI : FAILURE in 8h 52m 14s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/927069/4/20240929035217/NEC-FC : FAILURE in 9h 13m 13s","accounts_in_message":[],"_revision_number":4},{"id":"b2c48eb96b7225479d545234f47878a5e6fac45c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-17 14:10:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"38181eddc83cde8ae7ad348352b840e4e2029a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:10:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bdd3e34b402989e4cf099dce41fe66d72633363d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:30:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e3291d6c184c46b470f8f7b8d4657f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971e830da5b74695b27c7943aa592419 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac2e3d8b21340609f9c5380a09d6c43 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":4},{"id":"ac9731f8d6f7ee254aa94beacb42d8106a7be4a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:30:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6bee956ced81fc55ca6a621951aa3f0b6a7d2fc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 14:31:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c6e15312dc4425a48a43976072e61b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c315afc3df406aaba5f7180eaeefe8 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7330fd9b79d36350ca2ee139b2f9f060d6dca8fe","revisions":{"be23081eaef4dadf9df9d2e9b5e1cbc34ba10a5b":{"kind":"REWORK","_number":1,"created":"2024-08-26 04:29:14.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/69/927069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/927069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/927069/1"}}},"commit":{"parents":[{"commit":"788af30d09fe32a30c8c3a09de91af4e8d70cf8f","subject":"Merge \"[Pure Storage] Fix CG, FC WWN and NVME-TCP NSID bugs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/788af30d09fe32a30c8c3a09de91af4e8d70cf8f"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-26 04:21:05.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-26 04:21:05.000000000","tz":540},"subject":"Hitachi: Correction of Broken URL Links in the User Guide","message":"Hitachi: Correction of Broken URL Links in the User Guide\n\nThe URL links in the user guide were broken\ndue to the renewal of the document site, so they have been fixed.\n\nChange-Id: Ib103160250bb3d6f2e0a2fbb057601c31daa817a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be23081eaef4dadf9df9d2e9b5e1cbc34ba10a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be23081eaef4dadf9df9d2e9b5e1cbc34ba10a5b"}]},"branch":"refs/heads/master"},"9cfaca8fa5da4362954c66d75672f93a07a91f2c":{"kind":"REWORK","_number":2,"created":"2024-09-24 05:17:46.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/69/927069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/927069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/927069/2"}}},"commit":{"parents":[{"commit":"a14312cc34e49d9623846548719ec5472463b1f7","subject":"Merge \"Add cinder-manage command to update service_uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14312cc34e49d9623846548719ec5472463b1f7"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-24 05:14:39.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-24 05:14:39.000000000","tz":540},"subject":"Hitachi: Correction of Broken URL Links in the User Guide","message":"Hitachi: Correction of Broken URL Links in the User Guide\n\nThe URL links in the user guide were broken\ndue to the renewal of the document site, so they have been fixed.\n\nChange-Id: Ib103160250bb3d6f2e0a2fbb057601c31daa817a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cfaca8fa5da4362954c66d75672f93a07a91f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cfaca8fa5da4362954c66d75672f93a07a91f2c"}]},"branch":"refs/heads/master"},"9cdeb174d717db832f91dea99ba9fc28d30daa4b":{"kind":"REWORK","_number":3,"created":"2024-09-26 06:04:05.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/69/927069/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/927069/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/927069/3"}}},"commit":{"parents":[{"commit":"a14312cc34e49d9623846548719ec5472463b1f7","subject":"Merge \"Add cinder-manage command to update service_uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14312cc34e49d9623846548719ec5472463b1f7"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-24 05:14:39.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-26 06:04:05.000000000","tz":0},"subject":"Hitachi: Correction of Broken URL Links in the User Guide","message":"Hitachi: Correction of Broken URL Links in the User Guide\n\nThe URL links in the user guide were broken\ndue to the renewal of the document site, so they have been fixed.\n\nChange-Id: Ib103160250bb3d6f2e0a2fbb057601c31daa817a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cdeb174d717db832f91dea99ba9fc28d30daa4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9cdeb174d717db832f91dea99ba9fc28d30daa4b"}]},"branch":"refs/heads/master"},"7330fd9b79d36350ca2ee139b2f9f060d6dca8fe":{"kind":"REWORK","_number":4,"created":"2024-09-27 08:41:12.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/69/927069/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/927069/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/927069/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/927069/4"}}},"commit":{"parents":[{"commit":"a14312cc34e49d9623846548719ec5472463b1f7","subject":"Merge \"Add cinder-manage command to update service_uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14312cc34e49d9623846548719ec5472463b1f7"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-24 05:14:39.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-27 08:41:12.000000000","tz":0},"subject":"Hitachi: Correction of Broken URL Links in the User Guide","message":"Hitachi: Correction of Broken URL Links in the User Guide\n\nThe URL links in the user guide were broken\ndue to the renewal of the document site, so they have been fixed.\n\nChange-Id: Ib103160250bb3d6f2e0a2fbb057601c31daa817a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7330fd9b79d36350ca2ee139b2f9f060d6dca8fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7330fd9b79d36350ca2ee139b2f9f060d6dca8fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
