)]}'
{"id":"openstack%2Fcinder~927674","triplet_id":"openstack%2Fcinder~master~I5d4ed3a95f51daeef94dce83c4a9153f633008ba","project":"openstack/cinder","branch":"master","topic":"fix_misuse_assert_called_with","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2024-11-15 11:28:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d4ed3a95f51daeef94dce83c4a9153f633008ba","subject":"Fix a typo for assert_called_with","status":"MERGED","created":"2024-09-01 05:24:06.000000000","updated":"2024-11-15 11:28:10.000000000","submitted":"2024-11-15 11:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927674-fix_misuse_assert_called_with","meta_rev_id":"600d658eb6d6930d0695aa5727d58dd2d415645f","_number":927674,"virtual_id_number":927674,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 11:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-01 11:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-11-01 11:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":1,"date":"2024-11-01 11:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-01 11:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-15 10:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-15 10:02:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-01 05:24:14.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-01 05:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-01 06:06:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-01 06:12:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-09-01 06:24:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-01 06:42:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-01 07:34:49.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-09-01 08:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-09-01 08:26:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-01 11:16:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-02 02:58:52.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-02 14:40:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-03 03:16:41.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-09-03 03:17:23.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-09-05 06:12:57.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"REVIEWER"},{"updated":"2024-09-09 09:35:29.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-09-09 15:26:21.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-09-10 09:37:45.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-09-27 15:36:42.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-09-28 17:39:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-29 09:07:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-05 15:19:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-10-25 17:51:28.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-15 10:02:22.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"c1d4baccc6b97bb11b740382f31f8790f748ac09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-09-01 05:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2386f1918cf33f37a3c46263428e1d8a34576968","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-01 05:24:14.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/449/","accounts_in_message":[],"_revision_number":1},{"id":"510f68980f8d1b6beb88d1b9c950dc698aefc87b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-01 05:29:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"ca8d21af4f5b680f64d73c0892e848ac9b940ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-01 05:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6644b65cfcd4e1cb47075a3deafdca4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18062c134b0466397ae8a619c75a427 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef67df4425544396a5539821c01e94d2 : SUCCESS in 10m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9547240174c49bf83b9b9cb0d948c93 : SUCCESS in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc3bd81389a64b5697c00362a8de232f : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c5b9cf4ffc4ce499edcab9134d4e71 : SUCCESS in 9m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/011114cd31fa45b790c8d8d48ad13042 : SUCCESS in 12m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e82de56e71ee4e589ed81f505067d651 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"740da956863544bafb17a8bfb1c2d159da66a14e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-01 05:44:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab90326a17174036a3b8f6376b2f20e4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa7fe90c186a4a218448b7bba673095e : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/159e776ffdfe4a24a7a7aa6f8db603f7 : SUCCESS in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1917c7153b86bf7fbbab92b608b5e3bc4a33eaa4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-01 06:06:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927674 : FAILURE in 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"97c43841ba32c1ca252ffc80c3110cc9991a69ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-01 06:12:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/454/.","accounts_in_message":[],"_revision_number":1},{"id":"e1646e2036b12fccfea0b6664c0b59722669f9b5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-01 06:24:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927674 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":1},{"id":"40c3fde6817308570e64fd14efc5526949ff902d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-01 06:42:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3b7e9d6d8ad94e34b0d6f132bbfb068d : SUCCESS in 1h 13m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/be4d31239a3c41df804e36822c4da541 : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"caee791d8a9144c28e773a5772acc07b9b174f2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-01 06:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1042/.","accounts_in_message":[],"_revision_number":1},{"id":"4d17356905b2b7287056523a77afe4e3f7eff6eb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-01 07:34:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerStore_iSCSI/c5b204f/DellEMC_PowerStore_iSCSI/1692 : SUCCESS in 1h 20m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerStore_FC/40b252a/DellEMC_PowerStore_FC/1890 : SUCCESS in 1h 07m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerStore_NFS/6ad0d5c/DellEMC_PowerStore_NFS/1385 : SUCCESS in 1h 15m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerStore_NVMeTCP/e064a66/DellEMC_PowerStore_NVMeTCP/1245 : SUCCESS in 1h 54m 44s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerStore_NVMeFC/dc8eae3/DellEMC_PowerStore_NVMeFC/209 : SUCCESS in 1h 18m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"1894d7e24098036f8215d051f89ca6bb552ae6f3","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-01 08:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/927674/1/fc : SUCCESS in 0h 38m 36s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/927674/1/iscsi : SUCCESS in 0h 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9ebe61c805bb616060f4f312a2e50db4c2774f19","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-01 08:26:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/927674 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0220a7c42b5f51aed1dbe919505b3bfbf16b234b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-01 11:16:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerFlex_3.6/35bed09/DellEMC_PowerFlex_3.6/2020 : SUCCESS in 2h 17m 18s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerFlex_4.x/8484e48/DellEMC_PowerFlex_4.x/902 : SUCCESS in 1h 58m 34s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/8331131/DellEMC_PowerFlex_4.x_NVMeTCP/868 : FAILURE in 5h 52m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"8788e03a826e88986546de72f20186c7d1e30a09","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-02 02:58:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/1/20240902080437/NEC-V-ISCSI : SUCCESS in 3h 33m 35s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/1/20240902080437/NEC-V-FC : SUCCESS in 3h 54m 12s","accounts_in_message":[],"_revision_number":1},{"id":"570371376e91769561f8e735924f011eb1aea954","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-02 14:40:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/1/20240902175507/NEC-ISCSI : SUCCESS in 5h 21m 49s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/1/20240902175507/NEC-FC : SUCCESS in 5h 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"72242d62b0820bb189f532a335a31362c0be72f8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-03 03:16:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"244042880f997da664125371dd291faa5afade61","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-03 03:17:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da0d05903f86c7bc68102729309dfdd4380b24f2","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-09-05 06:12:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"edf2fd86690fb3a777556edeb7aacf9d2e0b76d8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-09 09:35:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/927674-1/HPMSA-iSCSI/ : SUCCESS in 3:18:59\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/927674-1/HPMSA-FC/ : FAILURE in 43:39.57","accounts_in_message":[],"_revision_number":1},{"id":"e3e0fd3be3569e34c2cb514946452cfe8f7ddf38","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-09-09 15:26:21.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"77d9a94553377d9e666ff10b9c8ee88e31c48612","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-09-10 09:37:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/927674-1/Seagate-iSCSI/ : SUCCESS in 2:23:52\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/927674-1/Seagate-FC/ : FAILURE in 2:26:51","accounts_in_message":[],"_revision_number":1},{"id":"919e45b2310c14c60180f22e6d8766242ed26011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-09-27 14:02:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"080ffedff6c1c86564808c6e1e3924125123dace","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-27 14:03:02.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/640/","accounts_in_message":[],"_revision_number":2},{"id":"de2c92ae18af1f43e5434e6e40696c59c59a9ffd","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-27 14:03:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f8f08a6a0535a72d80917f43521f26c38871f318","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-27 14:08:30.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"ab07f814bd06f4343dfac146a568fac1b840f27a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:22:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076233cfe7b3450184e1eaf330ee571c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a8ec41454dd4d41ac828ad37538022a : SUCCESS in 17m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9fc0f53af69143eabf77905fd9d0ae68 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bea3d6b7ae05c6abbd9e43d73014d59b74318985","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-27 14:50:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927674 : FAILURE in 46m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d4da659e090120503a3804e2814f544758b7dd28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 15:06:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/882578555c8c4aef8e3f53b4df35b873\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f41ba12c7049a68de100c57378c291 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8190837b6654b7fb2f4378676239f89 : SUCCESS in 11m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82a3330638454d71b4f0c96d5e5a4c1b : SUCCESS in 11m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52d4e3f55af242fc91b74512908e6067 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f793db6f65444b9eb3dc2f42b87eee : SUCCESS in 10m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a82d88cd9e5475fb9f6b49545576317 : SUCCESS in 1h 00m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b006ad4dc8d3415c9299436d7bd4390b : FAILURE in 14m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3049bc12dce493ea419c2c148a6392e : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a31cfd96fef9e78970e7722dee887563f4ace2c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-27 15:36:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/179982e7752341daa228b9ba870e6621 : SUCCESS in 1h 17m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f5d7696873c4df085bb52867db1ae32 : SUCCESS in 1h 31m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e47fb3f771af56dc0f522aa5974f90f759f89e90","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-09-27 15:36:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fdb8c1ba47dcff5ba16ba9ca7975b3c592237b3a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-27 16:06:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927674 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5f685dda4d2460457b0de596555428afc28b7561","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-27 19:16:04.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerStore_iSCSI/34359f2/DellEMC_PowerStore_iSCSI/1771 : SUCCESS in 1h 42m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerStore_FC/daba613/DellEMC_PowerStore_FC/1969 : SUCCESS in 1h 06m 21s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerStore_NFS/89c75cc/DellEMC_PowerStore_NFS/1464 : SUCCESS in 1h 27m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerStore_NVMeTCP/56194b4/DellEMC_PowerStore_NVMeTCP/1323 : SUCCESS in 2h 05m 27s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerStore_NVMeFC/7af60e9/DellEMC_PowerStore_NVMeFC/288 : SUCCESS in 1h 12m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"e8b64aaf400dacaf5ed8844346cfc0b44fa1e2e4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-28 12:00:09.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/927674/2/fc : SUCCESS in 0h 45m 53s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/927674/2/iscsi : SUCCESS in 0h 59m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bb50464502c23ddc0cc15a4d6f3bfbecef480222","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 17:39:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a7b052318b4e99424f5e2106e62970b33c5125c6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-29 07:51:17.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/2/20240929090531/NEC-V-FC : SUCCESS in 3h 56m 08s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/2/20240929090531/NEC-V-ISCSI : FAILURE in 7h 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3974a2baa34bfde287f715d780b4f67ebf0bfc90","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-29 09:07:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_Unity_FC/d1f92fb/DellEMC_Unity_FC/2662 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_Unity_iSCSI/3efd5c2/DellEMC_Unity_iSCSI/2671 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"83fc2b1d7050a7f84fc9e3cbc5612972a638612f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-29 10:14:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerFlex_3.6/abae73b/DellEMC_PowerFlex_3.6/2155 : SUCCESS in 2h 36m 20s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerFlex_4.x/8ce899c/DellEMC_PowerFlex_4.x/1039 : SUCCESS in 2h 03m 26s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/46140c6/DellEMC_PowerFlex_4.x_NVMeTCP/979 : FAILURE in 5h 54m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"08fe92ec4480a7a85bf456a966fac732d7f0c1b8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-30 02:36:03.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/2/20240930030808/NEC-ISCSI : SUCCESS in 5h 16m 44s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/2/20240930030808/NEC-FC : FAILURE in 8h 27m 51s","accounts_in_message":[],"_revision_number":2},{"id":"081d6209e68442260bf7775ef67caa3f6afeb723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-05 12:14:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"44988c0da9ef658b499881c7ebe78483517ffa62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-05 12:14:53.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/689/","accounts_in_message":[],"_revision_number":3},{"id":"a4945866c963bd8f25cad34f44a40f99aca7267f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-05 12:18:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"3a221eeaf002f90f878e41dd20ff6799a8b1a758","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-05 12:38:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/927674 : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":3},{"id":"04f4c87aa1f5c6ba08be693f84c7a3c57e62a5ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 12:41:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa77a73efe324998964e0d8068472b61\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c868467da4e047fc93b46416c4b3561d : FAILURE in 25m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32ed30c78c114fd99ea25049fb3813bb : FAILURE in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a7c30b036e5527ca4f8c162850355273aab8aa8b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-05 13:25:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/927674 : FAILURE in 1h 09m 21s","accounts_in_message":[],"_revision_number":3},{"id":"039912e9d152c8ac796040c59f0ac3a38f8830cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 13:27:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ece82857fb74901abe9998ca6c5d44e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ddf763bae34aefa59276cd70461744 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b354cfb90e9944099c5e537a71d9ff83 : FAILURE in 12m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c8a76d885594eb3b230fade418d136e : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84190309f74946f68f5d3d2555ed766d : SUCCESS in 7m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d76159fe3b94ce48ead327afb802210 : SUCCESS in 1h 06m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1052fa282184e90b316a1c34fb076ef : SUCCESS in 13m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe547b7778e64528bafeedc78414c9fc : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8dbcc174482715e0db7b1121537cdc95718366b5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-05 13:43:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf2dbe8c27554bc78e51c53cd839554c : FAILURE in 1h 15m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fd249506e50a4778b7e625046c6e6e09 : FAILURE in 1h 26m 24s","accounts_in_message":[],"_revision_number":3},{"id":"091043ea4d2ba69f9b4563356f1a0d6cc20fc928","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-05 14:23:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7654b46cb98b4fed467d758f2a59b88779ea9bfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 14:52:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c80527416454e39aab518f74c899984\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fd0f893bbb04f41952aeb8b210d6a93 : FAILURE in 25m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99ca7be307d0460cb39fc134800319a2 : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0762d5e05c4b83e062dc9bb34326507d29739ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 15:13:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c67a4ab3a6447a2b69ceba7aa4d4497\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2dcb26ec0a44f01a78e2983b6aae9b9 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f91249f66ef4ee88416852ad11ada9a : SUCCESS in 8m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d355086fc87a4edc994595d706642135 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075e80f2fd4e4b47b0c9c30034857175 : SUCCESS in 12m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73c52c4a854c4facb5233b7075695133 : SUCCESS in 49m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80236856741742b3a16f03f723517473 : FAILURE in 14m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d140bfc660a9438592dc77c2c888c134 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"2f948586bb5929c26461cf530f5937224525f8a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-05 15:19:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_10_2024_15_14_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"cfc8155ddca9b0d04308ad6f528db2cd4b3edd47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-05 15:50:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7c5cdc88e7b24aba89444230cc877e97 : FAILURE in 1h 18m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/687aa5d4d975451699bce17843d1bad9 : FAILURE in 1h 25m 08s","accounts_in_message":[],"_revision_number":3},{"id":"fa4f2e3e8ce948594e0e3f9c0808323af027383d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-05 18:04:25.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerStore_iSCSI/dff6b3e/DellEMC_PowerStore_iSCSI/1797 : SUCCESS in 1h 19m 11s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerStore_FC/679beab/DellEMC_PowerStore_FC/1995 : SUCCESS in 1h 01m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerStore_NFS/36ae858/DellEMC_PowerStore_NFS/1490 : SUCCESS in 1h 14m 34s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerStore_NVMeTCP/68aef31/DellEMC_PowerStore_NVMeTCP/1349 : FAILURE in 2h 50m 04s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerStore_NVMeFC/164d8de/DellEMC_PowerStore_NVMeFC/314 : SUCCESS in 1h 25m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"da2053aafdab40094232029c5411898f38cca5e6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-06 20:44:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/3/20241007015018/NEC-V-ISCSI : SUCCESS in 3h 43m 19s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/3/20241007015018/NEC-V-FC : SUCCESS in 3h 54m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8e10f654fa8fa55788f42e3166197ff0f808788a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 03:26:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerFlex_3.6/28f588e/DellEMC_PowerFlex_3.6/2192 : SUCCESS in 2h 17m 07s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerFlex_4.x/6eff44b/DellEMC_PowerFlex_4.x/1076 : FAILURE in 1h 58m 53s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/fe574fb/DellEMC_PowerFlex_4.x_NVMeTCP/1011 : FAILURE in 5h 54m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d14273173270e5569297339e225af6e85d50c24f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-07 07:46:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_Unity_FC/a538048/DellEMC_Unity_FC/2699 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/3/check/DellEMC_Unity_iSCSI/06bfdb6/DellEMC_Unity_iSCSI/2702 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"30197836147f68692583ccd2ed009c553e47bae5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-07 11:36:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"4c46479f3dba0eeb4fdd44f0d4efbed288a1aa2a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-07 17:02:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/3/20241007174010/NEC-ISCSI : SUCCESS in 5h 18m 25s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/3/20241007174010/NEC-FC : SUCCESS in 8h 22m 09s","accounts_in_message":[],"_revision_number":3},{"id":"d498e1a11038d075b52846ef478d6075a7ffe7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-19 12:54:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8edc89cb7b8848a61d0e9f508a8a7b9812889fc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-19 12:54:52.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/774/","accounts_in_message":[],"_revision_number":4},{"id":"ff3c29cb2fa5d69f6c1d02d789f9effaf4160f12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-19 13:01:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"fababf2f041143c17a5bf915df453d5af5a42a2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 13:13:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67140e7f81147f6b169fbc37db65c5e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b40cd1074e754135a56ec2dbf55144df : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c64f0571b4094718b523b510832d087e : SUCCESS in 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"faec80185fde0895fa3a7b76b2877c00c14a0639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 13:52:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4077fefe11dd4af9ac683ca69e294ac4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a714d38ff7463a8c06f1a2ea888e1f : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023268454ffc4dd082ff5c1dc7ad3851 : FAILURE in 11m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6c32cd6d8b9473d91ccf07a33885b51 : SUCCESS in 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681172e0a5b8482289870cf43502810b : SUCCESS in 12m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff6b5fcb2fd94c338dcf2b4896116b38 : SUCCESS in 51m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d254c0cedc184e0fa15b697f9df80c34 : SUCCESS in 15m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ec8a705d471483b93dc27d421ae6b55 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"33d8e38662324e1fe5389a868012ffb478e85120","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-19 14:26:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2e7eeeceda9b4468b5b337243863aa56 : SUCCESS in 1h 23m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f5e8a86cbbec4e5cac97b475a1877749 : SUCCESS in 1h 29m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7a13f09a61cdcf516f7c358c40ba099c3b258d14","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-20 19:28:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/4/20241021004218/NEC-V-ISCSI : SUCCESS in 3h 32m 03s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/4/20241021004218/NEC-V-FC : SUCCESS in 3h 46m 38s","accounts_in_message":[],"_revision_number":4},{"id":"a6b5948755cca3575de599f6b429278256b04115","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-20 23:40:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/4/check/DellEMC_PowerFlex_3.6/ff8b211/DellEMC_PowerFlex_3.6/2265 : SUCCESS in 2h 17m 26s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/4/check/DellEMC_PowerFlex_4.x/ceaab31/DellEMC_PowerFlex_4.x/1148 : SUCCESS in 1h 56m 42s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/ceb8c99/DellEMC_PowerFlex_4.x_NVMeTCP/1069 : FAILURE in 5h 52m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5de720f147a90f21f67667415535b4901e8378e8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-21 01:52:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/74/927674/4/check/DellEMC_Unity_FC/e556dc9/DellEMC_Unity_FC/2765 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/4/check/DellEMC_Unity_iSCSI/b47bb60/DellEMC_Unity_iSCSI/2756 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3a20d87b9d62af282e433a68e461b9bc0315303c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-22 01:13:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/4/20241022014914/NEC-ISCSI : SUCCESS in 5h 16m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/4/20241022014914/NEC-FC : SUCCESS in 8h 24m 11s","accounts_in_message":[],"_revision_number":4},{"id":"bcc321a1228a2ab5a448a577936ff482abc36a0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-25 17:14:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7c6ae89b715075135a582f8d1975a44b751e6cda","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-25 17:14:49.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/810/","accounts_in_message":[],"_revision_number":5},{"id":"238a1745860b88619395366974dcaccb14a7e4a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-25 17:21:21.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"3d3e5712275a74a4522a13ff37a8ffd910854051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 17:36:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5704438c78ba4ff4b3ca2d23de348d70\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d594a76a0c2b4645afa3d701baa0df4d : SUCCESS in 17m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cdec6ac62da14a30aa0281a3cce31c65 : SUCCESS in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"20dce4fef7a3166608206115839ecd643a5efcde","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-25 17:51:28.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/4d11a09a82de48308fbdb4a5b8bf6391\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/45388b570bf44f7b9ed2325462661e5d : RETRY_LIMIT in 5m 50s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/f45d24dee5494088a61150e4090628ae : RETRY_LIMIT in 5m 36s","accounts_in_message":[],"_revision_number":5},{"id":"ecc05761686cd4fac8c2c5eb73b66f1b59e352d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:17:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab0588dbd4c84e04817176ecffaa4ff7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ecb90c93c54b58b2971d942c3d2fbc : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07748993745841ab9294cb0c7d7cf4b5 : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cac1f88da7e4fc88877dea54696c09a : FAILURE in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d764d0d9e9d84232b643ed1a2caea92b : SUCCESS in 13m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af6c46088cce430291b052316f163d6c : SUCCESS in 59m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d5aad9bccdd487eada3665a8c74a6c6 : SUCCESS in 17m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/194dd35b9818410dab1cede08bb2ce9c : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b1490ccd836caa9e332d01f623bf741832c35222","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-25 19:09:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b7d0e8ddb0a6442abf77bc0656738092 : FAILURE in 1h 14m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/84cd2cea46004ccea53aa16d9452f995 : SUCCESS in 1h 18m 25s","accounts_in_message":[],"_revision_number":5},{"id":"817ea0ae26178540d9d5f474bca1cafbe87042bb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-27 08:40:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/5/20241027133655/NEC-V-ISCSI : SUCCESS in 3h 53m 55s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/5/20241027133655/NEC-V-FC : SUCCESS in 4h 03m 50s","accounts_in_message":[],"_revision_number":5},{"id":"46d8f751143dce54e93e874bbb984cb812c41bbf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-27 17:54:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/5/check/DellEMC_PowerFlex_3.6/785f740/DellEMC_PowerFlex_3.6/2304 : SUCCESS in 2h 45m 13s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/5/check/DellEMC_PowerFlex_4.x/226f310/DellEMC_PowerFlex_4.x/1184 : SUCCESS in 1h 59m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/73bd584/DellEMC_PowerFlex_4.x_NVMeTCP/1097 : FAILURE in 5h 54m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"e8ed74412f4da55b2424785edb835f7d5bd1b1d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-27 20:08:23.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/74/927674/5/check/DellEMC_Unity_FC/9268173/DellEMC_Unity_FC/2798 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/5/check/DellEMC_Unity_iSCSI/37e8eca/DellEMC_Unity_iSCSI/2782 : ABORTED in 6h 00m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"966b46449645141a2f806dea9f37878c17f1a2e3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-28 08:07:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/5/20241028110901/NEC-ISCSI : SUCCESS in 5h 16m 32s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/5/20241028110901/NEC-FC : SUCCESS in 5h 58m 21s","accounts_in_message":[],"_revision_number":5},{"id":"b0bb0dd36841d7d2b88f1fe4ffb14380876fccfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-11-01 11:09:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6683eb62dffef9477247098251cc1ec662107e19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-01 11:09:13.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/843/","accounts_in_message":[],"_revision_number":6},{"id":"ced189451205e900483032e164bb935db390031e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-01 11:16:05.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/927674/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"75c875ef435963c9b670e2c9f5af5f96c25a5358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 11:43:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c21f16e92cd4d6aae62a6874b798d1b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50887f3daeed427d806dcf9d7d33ead3 : FAILURE in 30m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84d449dbb85d4f6f95690f8810162780 : SUCCESS in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9e93b6eca56d6346c9f71a49c8237fc5ff4827cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 12:19:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f256000a0d284d88a98ce18cf2293794\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/346cd543aa5a474baef3705d4ef979ac : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e61d571b06c4809847c1e1f2fc40d2c : SUCCESS in 13m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82c057d37a6544f79373ee03ef79dab4 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ab6fc516b04cfd9415effe32573562 : SUCCESS in 11m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/683dcc6834734240aa3cbd588dd837a1 : SUCCESS in 1h 04m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/896e1ae685614f959aabd47f7e9ab2e9 : SUCCESS in 17m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3c4a8dddc293400987d7321f5dd256ad : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":6},{"id":"a25d037f0cc8f75de06633ce85d3a7ba6cfe01bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-01 13:04:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/794/.","accounts_in_message":[],"_revision_number":6},{"id":"01095ac0ad4417557d7c7284e62e600e2a05ee15","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-01 13:49:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4cc6ab9deee400ca2f88faf1c1ddf6d : SUCCESS in 1h 13m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a7e8849217e8498b9068a21977f64887 : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":6},{"id":"f4772aca0b5e78e690e6c15ea75fb432fa1bfbc1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-01 14:30:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1365/.","accounts_in_message":[],"_revision_number":6},{"id":"f17527d974d559f91562c42a7a89e930c2354561","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-01 17:33:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/927674-6/Seagate-iSCSI/ : SUCCESS in 2:21:44\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/927674-6/Seagate-FC/ : SUCCESS in 2:15:57","accounts_in_message":[],"_revision_number":6},{"id":"0691395fbb5a4f77811f2c6166e9f83a6bf5f74d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-01 18:43:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/6/20241101235613/NEC-V-FC : SUCCESS in 3h 40m 10s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/6/20241101235613/NEC-V-ISCSI : SUCCESS in 3h 47m 31s","accounts_in_message":[],"_revision_number":6},{"id":"a28fa537f41b47a9fbc5996f193bbfd8d5afd6f9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-01 20:20:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/74/927674/6/check/DellEMC_PowerFlex_3.6/853261b/DellEMC_PowerFlex_3.6/2336 : SUCCESS in 2h 21m 37s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/74/927674/6/check/DellEMC_PowerFlex_4.x/c4ba067/DellEMC_PowerFlex_4.x/1209 : SUCCESS in 1h 59m 09s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/74/927674/6/check/DellEMC_PowerFlex_4.x_NVMeTCP/263269b/DellEMC_PowerFlex_4.x_NVMeTCP/1115 : FAILURE in 5h 54m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"a9058df933e456a2b9ea449583c5bc1df9d076c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-01 20:33:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/74/927674/6/check/DellEMC_Unity_FC/8158a06/DellEMC_Unity_FC/2821 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/74/927674/6/check/DellEMC_Unity_iSCSI/7ced0c8/DellEMC_Unity_iSCSI/2799 : ABORTED in 6h 00m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"d0def9dcdd38ee7978ff28b021d78bf1b130dd6b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-02 06:27:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/6/20241102093759/NEC-ISCSI : SUCCESS in 5h 22m 16s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/74/927674/6/20241102093759/NEC-FC : SUCCESS in 5h 49m 26s","accounts_in_message":[],"_revision_number":6},{"id":"f48b6dfe9df381fa4c87abbf1462d73de2e7bf9a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-11-15 10:02:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"55955b21345e3724aadd1e52aa65aa16bd343c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 10:02:54.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"578e541edd1d6cee7114f0527db5540db6fa10e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:28:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2746350e2f48b0ac628518582cd7f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7f742d7320408aa5c3b1a1c7639f7f : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2834a8afc343078ff7420a93cbeb49 : SUCCESS in 18m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dd9d0484bc942ff96af56d72672a317 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff20aad8b7ff453eb9bf817ab9cb7639 : SUCCESS in 13m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d46c5a6abed640d0bc8660ee175f05bb : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":6},{"id":"600d658eb6d6930d0695aa5727d58dd2d415645f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:28:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cf3205ecf88518e557d3b35fb9588730914ab19f","revisions":{"d27f1df9ee1a50539c13e34e7083198559d6508a":{"kind":"REWORK","_number":1,"created":"2024-09-01 05:24:06.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/1"}}},"commit":{"parents":[{"commit":"9fbe7f7046d7dd6c0aa1e4af428d112e4d6d80e5","subject":"Merge \"Add API document for availability zone list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fbe7f7046d7dd6c0aa1e4af428d112e4d6d80e5"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:22:31.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d27f1df9ee1a50539c13e34e7083198559d6508a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d27f1df9ee1a50539c13e34e7083198559d6508a"}]},"branch":"refs/heads/master"},"c5f699e527550f794988cbf5cfb3f31d22f083a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-27 14:02:54.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/2"}}},"commit":{"parents":[{"commit":"b07e4147f9c3ad2851dae4bb75a80a5a2c7f1157","subject":"Merge \"Ceph: Add option to keep only last n snapshots per backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b07e4147f9c3ad2851dae4bb75a80a5a2c7f1157"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-27 14:02:54.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5f699e527550f794988cbf5cfb3f31d22f083a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5f699e527550f794988cbf5cfb3f31d22f083a2"}]},"branch":"refs/heads/master","description":"Rebase"},"6bdc49a16de01fb86a05f4e3cc76be74d55955d7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-05 12:14:45.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/3"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-05 12:14:45.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bdc49a16de01fb86a05f4e3cc76be74d55955d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bdc49a16de01fb86a05f4e3cc76be74d55955d7"}]},"branch":"refs/heads/master","description":"Rebase"},"e8060c7b847d905907136a453274fe4a0e32d93f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-19 12:54:42.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/4"}}},"commit":{"parents":[{"commit":"a357d009fad1ca6c685861acab85b6d0d74f309f","subject":"Merge \"Fix mypy errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a357d009fad1ca6c685861acab85b6d0d74f309f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-19 12:54:42.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8060c7b847d905907136a453274fe4a0e32d93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8060c7b847d905907136a453274fe4a0e32d93f"}]},"branch":"refs/heads/master","description":"Rebase"},"a30d22dc14552682b1f70982576133031c38bd1e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-25 17:14:38.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/5"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-10-25 17:14:38.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a30d22dc14552682b1f70982576133031c38bd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a30d22dc14552682b1f70982576133031c38bd1e"}]},"branch":"refs/heads/master","description":"Rebase"},"cf3205ecf88518e557d3b35fb9588730914ab19f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-01 11:09:03.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/927674/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/927674/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/927674/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/927674/6"}}},"commit":{"parents":[{"commit":"c4bedc1de1d8dd0a00abb6f036fa8add0e72f437","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4bedc1de1d8dd0a00abb6f036fa8add0e72f437"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-09-01 05:20:55.000000000","tz":0},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2024-11-01 11:09:03.000000000","tz":0},"subject":"Fix a typo for assert_called_with","message":"Fix a typo for assert_called_with\n\nChange-Id: I5d4ed3a95f51daeef94dce83c4a9153f633008ba\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf3205ecf88518e557d3b35fb9588730914ab19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf3205ecf88518e557d3b35fb9588730914ab19f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
