)]}'
{"id":"openstack%2Fcinder~928846","triplet_id":"openstack%2Fcinder~master~I69f0469b1a6c543752502f2f915001caf4570686","project":"openstack/cinder","branch":"master","topic":"bug-2107328","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-11-03 15:51:10.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-08-04 14:52:51.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-30 07:20:03.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-18 19:53:02.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-02-18 21:50:27.000000000","reason":"Some votes were not copied to the current patch set"}},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-08-18 12:54:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}}},"hashtags":[],"change_id":"I69f0469b1a6c543752502f2f915001caf4570686","subject":"Add backups summary API","status":"NEW","created":"2024-09-10 17:51:05.000000000","updated":"2026-08-18 12:54:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":45,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1f8aa87332e3e2b3e2f2902c8bc54aba5997eff","_number":928846,"virtual_id_number":928846,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-04 14:52:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-08-18 12:54:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 17:51:16.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-10 18:31:33.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-10 18:42:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-09-10 18:44:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-10 18:55:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-10 19:12:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-10 19:25:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-10 20:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 02:59:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-11 12:42:01.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-09-11 16:53:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-09-11 17:43:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-11 19:06:29.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-09-11 22:51:49.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-12 01:20:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-12 09:53:47.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-29 15:13:12.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-09-29 18:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-10-02 14:40:20.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-10-04 04:54:19.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-04 15:53:29.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-04-04 07:14:43.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2025-04-14 15:00:23.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"state":"CC"},{"updated":"2025-04-14 15:00:23.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2025-04-14 15:00:23.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"state":"CC"},{"updated":"2025-04-14 18:06:42.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-04-18 18:11:51.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-22 13:20:59.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-07 15:47:57.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-23 00:09:44.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2025-10-24 05:54:53.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2025-12-18 10:21:17.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-18 21:49:52.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-02-18 22:18:47.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-19 08:55:10.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2026-02-20 14:49:05.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-20 14:53:29.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-21 20:06:42.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-06-12 22:41:34.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-12 22:56:59.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-15 14:35:04.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-03 09:47:27.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-03 09:47:39.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"}],"messages":[{"id":"7c8373f10543d89ce24d1a7544c9e39527558f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-10 17:51:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efab9f9a5032f12913e842d130962b6098f49775","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-10 17:51:16.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/519/","accounts_in_message":[],"_revision_number":1},{"id":"19ef4be581b4cc29b1a0c6e28443ad5adbc94a3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-10 17:51:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"68d738c59e33af68b8a115147efe46e06f6fc5e6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-10 18:31:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/928846 : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"02d1a0f2cc176c59595cd9e6220263efb0173359","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-10 18:42:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/512/.","accounts_in_message":[],"_revision_number":1},{"id":"0264d55f2251661b0370e6cdf1e6c55db90609bc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-10 18:44:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/928846 : FAILURE in 45m 28s","accounts_in_message":[],"_revision_number":1},{"id":"627e873ccfa4732d536989da2448f7b0352a0c18","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-10 18:55:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":1},{"id":"663c3def58e3c92ea6879ee782e83f9db7a002ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-10 19:12:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b80d90ff37cd4757abbce261bcb14d5d : FAILURE in 1h 08m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2dd0f1ff77604b238d200b0b8545251c : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6ab91f6674a7bd5131c68568110a4c745e173a0a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-10 19:25:39.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ed6e95c5aa4344b1bdb6366dcb5bbb4f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aeecc31092a940d296d5a5ef1df24571 : SUCCESS in 46m 35s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e39af0fe4364b67a77d209c4eed312f : FAILURE in 36m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24c5b46e92c94ce1a42fd247c3c3f538 : FAILURE in 1h 34m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/769bd528b99349778a2b3af45cbb0de7 : SUCCESS in 39m 48s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"62a1331b9a1b00562ef7faa2642fb348552d769e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 20:17:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a865b9f22f94fb8993fbfede2164e0a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbac233bcbbd4c4cbd229c05771c91e0 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ddc21b2fe24a14bf92b7e19089c1e1 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed7cb449b5664a019b1c2d81e60aea73 : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ab621e6f5004d0391f2f3b8c4483a7d : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce74a92802a04edf859076fe15a6c558 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9ea908659c4ad98b1da2f9774bba3b : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/0a08682eb0924b539e0feca14426f16e : SUCCESS in 49m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6254468afb1448a089f9ba94e69fc457 : SUCCESS in 1h 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68760dc45723447097f08fcedcaf8995 : SUCCESS in 1h 08m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/151e0274271044f8852acce615b920cb : SUCCESS in 28m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c374c5c58cc34f8b827c9467b555ed35 : SUCCESS in 17m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47f8768a8fc04424adbe516c5da70eeb : SUCCESS in 4m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4947e94daffc490e915e34c22bf55c70 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13016044a5644bc08f141206cc8d458b : SUCCESS in 5m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ff01ca7e5af46dd8350c1a226e578c5 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/815414f532de4738a3a652bcb8fb2943 : FAILURE in 26m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30fb6b1b35494c7dbf37cb6c68ad0a72 : SUCCESS in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/749f64273dcb48c7bf6a09c964a98394 : SUCCESS in 1h 54m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee3f2f09ac9742b6ad1028abc82bf32a : FAILURE in 2h 14m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e15d433dea3941ca9c4631665ed6a676 : SUCCESS in 1h 40m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d41feb952e51485f8b431866b3c5cb3f : SUCCESS in 1h 45m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/979a9b77eefc40789991c01d5913780e : SUCCESS in 22m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49ca5b00e4e34a598eb4045de0084cd6 : SUCCESS in 1h 16m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9b079339c7c84d6c8d5bd25e4f2dd6a7 : SUCCESS in 1h 56m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/db58d803accd49e98fd0e3a73d729b93 : SUCCESS in 43m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/faf66f22f50c44d2868edbb747866fca : SUCCESS in 1h 35m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5f756c7edae744699a03dbed4568ede3 : SUCCESS in 1h 33m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/824f868c018d4110a7ce2b3e18bf6d4f : FAILURE in 1h 38m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b288ec461d247508a7983aecd06d0ef : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ee6f8b70a6fde4967008e9ad2f30222db5b25971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-10 20:58:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b60dd17fcd6a5071da94c979e3f1b57df9306cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-10 20:59:01.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/521/","accounts_in_message":[],"_revision_number":2},{"id":"3a5c0f21f760c4840220d6abe7fc0f5452d12771","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-10 20:59:03.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"cfa44c2e96b9fd128c69989e65821e047603c448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 21:29:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d03a69bda134303a34a131cd5619c83\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/918ba0ec0a844e80af08c07e1ec78274 : FAILURE in 26m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/daab576bc21c42da9c0cb79071be0ccb : SUCCESS in 28m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"484bdf4a8d9636062c5494ac084c004895ab3a13","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-10 21:59:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/928846 : FAILURE in 41m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d1046bbe862caa3cd2ddf5a97ba6a1849310e11e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-10 22:03:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/928846 : FAILURE in 43m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c5ca322b7b3e4af1323d52c67a312c0dd846af4d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-10 22:30:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/562ca4ac1dfc4fdb94e300478d2ddeb6\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36929862fa324730bf682bb64df1ebbc : SUCCESS in 45m 39s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9961277d9c364a14bc53c28f8bf0e447 : SUCCESS in 42m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d64c860037c74885a1d34b80860cd0c2 : FAILURE in 1h 30m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7afd73da1bd245ef8b083a208ad8aa5d : SUCCESS in 40m 28s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"2dc1317c89469734ba1119bb39159494b7a7c5c7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-10 22:32:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/514/.","accounts_in_message":[],"_revision_number":2},{"id":"037220a9a282a258681e71aa21df31f23fcd91f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-10 22:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":2},{"id":"abe600606a11a4d33f14c7d92558f9e096a70541","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-10 23:29:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce23aca5525b498b9fa18a0da7b6ce95 : FAILURE in 1h 10m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/62e5377b67084f9f94481fa67d8aa315 : SUCCESS in 1h 19m 26s","accounts_in_message":[],"_revision_number":2},{"id":"de0b28efa8e7998949b741d78071a7aee28d047d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 23:37:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5163d98881b24f72b43a2980fbf67071\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3041d9d8e004d27ac38a174f8d1fd03 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07ac4d8da6541aa80f4742a472f3651 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/739957db715c45e88eebd01414d08f6b : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fa8d3cbf5554348a2157c8807557eea : SUCCESS in 9m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95b94f5a488540bfb509bd4df96663b8 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd07fedaa57e48c48ac67d9b334b3bf5 : SUCCESS in 10m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/da4e150c606442d29f5b3b38bfb2061f : SUCCESS in 54m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a1c52785d924e9da787516c17a5beca : SUCCESS in 1h 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf05bed4acb48f39405a92e496c300b : SUCCESS in 58m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c1dcff7eb1240c3bb730a5dd15d3456 : SUCCESS in 26m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d9bb94d8b924670ab2c32a54bb21764 : SUCCESS in 17m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8bf017423b6043929bcf3db44c8196b1 : SUCCESS in 4m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5607ddf25b33416092c1d9a3571a4ef9 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61f2798d45004dbc821e9fe75e37f4ac : SUCCESS in 6m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c5446370c80246e9988044bcd69f7a40 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/43bae7c041544d4589951ea7b09c8ed9 : FAILURE in 39m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3cd394b09b24910ac63f0470b403cdf : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a415c0db414b43a38eb881d885698dc7 : SUCCESS in 1h 24m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a033185256a47729d60c41ea266cd7a : FAILURE in 2h 32m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0cde69a8743447898c7dc56780dcd863 : FAILURE in 1h 14m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9899c7a92fcf4156a366322a438a601b : SUCCESS in 1h 30m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e673b8c6f3541d0adfb4434d18ecc9a : SUCCESS in 37m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d59e8c185ed741e0bee441f62febd0aa : SUCCESS in 46m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/69cf05ae64a8453cbd52e15ca7552e7b : SUCCESS in 1h 32m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2ffafa2aae4c402483c23b28acce4f51 : SUCCESS in 49m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5ee3b828da8049c58e68b872e871b07c : SUCCESS in 1h 30m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a0ccdae2caf14644a8bd450a978046b7 : POST_FAILURE in 2h 02m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c172181ce6f54a8c837fb95e35b45d3b : SUCCESS in 1h 24m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07273f642415488486a84b3d9f34b17c : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a6847a8ca4d118df416492118802ccc8f9dd8c0a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-11 00:17:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1103/.","accounts_in_message":[],"_revision_number":2},{"id":"afa8a92c9bcadd41ac949b842922da44523670df","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-11 02:59:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_Unity_FC/fa32b1e/DellEMC_Unity_FC/2583 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_Unity_iSCSI/ae2c51b/DellEMC_Unity_iSCSI/2588 : SUCCESS in 2h 01m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fffe1ed96cff36f1ae778ef8663a793dbc195501","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:29:10.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/522/","accounts_in_message":[],"_revision_number":2},{"id":"5e6454bc3b91eb5a0333d319ffba9369bf895545","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:29:17.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/fc.log : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"bb7caa08d13af23009bd821b10787f2cec353460","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:29:21.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/523/","accounts_in_message":[],"_revision_number":2},{"id":"14c03e53d9b554ae7403ea5c3781144f9731833d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:31:14.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"da81d652627008c70499dd9b0f13d503306131a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:36:26.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/524/","accounts_in_message":[],"_revision_number":2},{"id":"889adc1bca7720a57732e2bcf4e2143d64fc7193","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:41:19.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"8027cde21c812ca75b356d22fd1cbaf357c56e88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:50:43.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/525/","accounts_in_message":[],"_revision_number":2},{"id":"6b8de327153b5637fd0dd5e960bbecf3ac674c59","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 07:55:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"bbaca680eb05fc52bbc2074009e65a36af8ed719","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 08:09:14.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/526/","accounts_in_message":[],"_revision_number":2},{"id":"18f2868a568759240dea46f2907bb3b52128ab95","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 08:13:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"68653a8a6e2f08b7d14c0e1018f8be7cdcc4cec8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 09:48:07.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/527/","accounts_in_message":[],"_revision_number":2},{"id":"bcee47494a7724217f28964859a300600bbde4c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 09:53:12.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"843f20c47ff07c967521efb9748d0045fe7ea85b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-11 10:04:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"a6cecdeb2fc8d08ffa4a07f094fb35a459f50482","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-11 12:42:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bcdb9bb3685f278855a8badfe10aee46c7a9005","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-11 12:54:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f216e22abdc5331c9fd4b5482b56a23819974f37","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-11 13:32:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/928846 : FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3d0d36e1562bac13158a701d09035b75d44e49c1","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-11 13:37:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/928846 : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":2},{"id":"bfa732501267399fd11d6999124e59fd180a30db","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-11 13:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f66a12edaaaae44a87651737a7100f99912d1235","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-11 16:11:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/243ce1ec6465415ebf18fbcc1be173ec : SUCCESS in 1h 14m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8dffd9f7e1974063b9093a45d510353f : FAILURE in 1h 17m 27s","accounts_in_message":[],"_revision_number":2},{"id":"32ac1dc866f0d6c4e1912abc1e02a37fb8468034","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:30:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29671795751d4937b376bcbb4432dfd6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21328ebd4f7a4cc19158b40d22b661d0 : FAILURE in 26m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa90ec7d02d5425386b4e9112161eaad : FAILURE in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c954abc5db6f33dc1faaaaa56e23c549c22bd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:48:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce9eeba2b3af47bd9e76f3870fd9b6a2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1e968fe25f354675a03ce030227d7532 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee17a58a3e549c58249ac2f957907b4 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77a8b49e5a774ff4817a4e9ab7fed3d4 : SUCCESS in 11m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/beffec8751a04097bd61a891e90dae11 : SUCCESS in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cb2da8fc907413b96fe876fdc59c3f1 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e038206a5ec449a8d75143f68bc85fd : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0c68d0dcf4ed430ea15e7e473cfcd3d9 : SUCCESS in 45m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc7e96b503c0439cac1b09c7cdecdc3b : SUCCESS in 1h 45m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/702d2d2b8d6844f092579f0e411df6c7 : SUCCESS in 1h 01m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d218b411002344cc981c1d0ed268766f : SUCCESS in 18m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b0962070bf444b7acf6221ed80f39ae : SUCCESS in 16m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ba8c2b84dc554cf49f3f8601c3ba2568 : SUCCESS in 4m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37f2478b8a9a4bc780bc85bfd13a0a34 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2b3f0794d3b479bbef61f43d3c734f4 : SUCCESS in 5m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d6a990281ab743ddb0ea00001e5b7cc4 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d4fe76b188e4e51a8eb12b5ed031de1 : FAILURE in 44m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83b1acd9aa494b00a281667e4e417063 : SUCCESS in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/401e4d95f0e145f1946868f7a6488624 : SUCCESS in 1h 50m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a5a2ad21ccba414ea86aa1c630b69d0c : FAILURE in 2h 51m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8548e8e21fda41f8862b2f9daf8e3a59 : SUCCESS in 1h 51m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d9f1854113724478b0376871756e983f : SUCCESS in 2h 19m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8380b77cd4404a43a600817bdd88cf1b : SUCCESS in 38m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8916963087945a5b0a7f032f5d97863 : SUCCESS in 1h 13m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd41257951de43e39deccb33c563a976 : SUCCESS in 2h 02m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/23c93c22782c4b40b549cc214087e7ec : SUCCESS in 28m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/761029f54e9f4b09a8bc0d491e591203 : SUCCESS in 1h 25m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cae3f9755e534167885cb9e649317a60 : FAILURE in 1h 18m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/300e2845ee2e4e1ba2df33cfd86d8c6b : SUCCESS in 1h 51m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1ea3eac727459cadfa8a565b4effda : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3d47cd03ca55025a6f0b23fca38251ebdd2359f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-11 16:53:34.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_09_2024_16_48_36\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2884754aa3b8efa403d00a57543c41a62097e201","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-11 17:43:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/928846/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/d76bc95/ : SUCCESS in 48m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/928846/2/upstream-check/cinder-tempest-plugin-ontap-nfs/73ea6e9/ : SUCCESS in 53m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c40639cc5d42a42f7ec157ea1a26c3f7dec42913","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-11 19:06:29.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerStore_iSCSI/96502f4/DellEMC_PowerStore_iSCSI/1729 : SUCCESS in 1h 24m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerStore_FC/8deff4a/DellEMC_PowerStore_FC/1927 : SUCCESS in 1h 06m 29s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerStore_NFS/a46a626/DellEMC_PowerStore_NFS/1422 : SUCCESS in 1h 22m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerStore_NVMeTCP/97539bf/DellEMC_PowerStore_NVMeTCP/1282 : SUCCESS in 2h 17m 44s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerStore_NVMeFC/e8e9f6f/DellEMC_PowerStore_NVMeFC/246 : SUCCESS in 1h 08m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"24dddab90ac2bfdb6697717501c4d986709370ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-11 22:51:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/2/20240912035457/NEC-V-ISCSI : SUCCESS in 3h 42m 19s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/2/20240912035457/NEC-V-FC : SUCCESS in 3h 56m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5c2c22a011e015f61a2c497cab56aae2502f1858","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-12 01:20:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerFlex_3.6/b4f23a0/DellEMC_PowerFlex_3.6/2077 : SUCCESS in 2h 13m 14s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerFlex_4.x/b89393e/DellEMC_PowerFlex_4.x/960 : SUCCESS in 2h 06m 18s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/0947c0e/DellEMC_PowerFlex_4.x_NVMeTCP/911 : FAILURE in 5h 53m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"2bff6e57c6716564f56ec49bb57b2372d7f139cb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-12 01:34:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_Unity_FC/cb1a10f/DellEMC_Unity_FC/2588 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/46/928846/2/check/DellEMC_Unity_iSCSI/89ebc01/DellEMC_Unity_iSCSI/2593 : SUCCESS in 2h 10m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2773dae0e3a575e63ec1956a174d121ce14690d9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-12 09:53:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/2/20240912102009/NEC-ISCSI : SUCCESS in 5h 20m 57s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/2/20240912102009/NEC-FC : SUCCESS in 8h 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2c86cc2778d1526589b275da828320d2938c47a6","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-29 15:13:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/928846-2/HPMSA-iSCSI/ : SUCCESS in 3:16:58\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/928846-2/HPMSA-FC/ : FAILURE in 25:42:15","accounts_in_message":[],"_revision_number":2},{"id":"994de1160171675312f20b356d6067b639addc98","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-29 18:00:09.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/46/928846/2/fc : SUCCESS in 0h 55m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/46/928846/2/iscsi : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0c9698a6da2b7a387a8c4898b53feec472305beb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-04 04:54:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/928846-2/Lenovo-iSCSI/ : SUCCESS in 2:25:26\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/928846-2/Lenovo-FC/ : FAILURE in 6:20:13","accounts_in_message":[],"_revision_number":2},{"id":"f9287cfdce79887266d08012fa19e5267814de11","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-04 15:53:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/928846-2/Seagate-iSCSI/ : SUCCESS in 2:23:06\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/928846-2/Seagate-FC/ : FAILURE in 6:19:35","accounts_in_message":[],"_revision_number":2},{"id":"3dbf3c707a9a50384b8e48864fccec8908589f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 11:00:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d248cb2b94a2856cf4573ffc8368189ec891cc55","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 11:00:38.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1695/","accounts_in_message":[],"_revision_number":3},{"id":"85336899902f0374948147349386e6358b21f133","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 11:00:40.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"0f8b3ae2765d2360efb68ce490bd7806ecfaa1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 11:01:08.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"266f571c184f8ac41bf00870e3e8d9df36945a41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 11:01:14.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1696/","accounts_in_message":[],"_revision_number":4},{"id":"db0853a4758822340eec1b88b422ad0605b9213c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 11:01:17.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"78836ab90d38ca9119a12bcc87563603f14d3071","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-14 11:02:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4fa35f1cbd79096feb5daa33b33daa7f0ac14604","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 11:35:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9843d9f125a4542907ca17c02a908bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4847bc298bd4b1a839f9b37d71f955b : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/554b8dfd46394992bd8d8a1ffc59b0c2 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd4f1c6cf82a4627b8e5b25146603377 : FAILURE in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"424ddbe068494ff46934a19862404f6ba789437d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-14 12:42:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e81ea95d703465aae36950609116b1d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/51c11db45c8b48798a0b522dd722bcf7 : SUCCESS in 42m 49s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6c23d5cead4486d83d7700d44ccba1d : SUCCESS in 41m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7ebf941492574df9a3d689e50f2c5093 : FAILURE in 1h 38m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8bbdb763791046dbae48373e14dff0e5 : SUCCESS in 46m 34s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a459f23e1f54d2c9108c074c89a0ae9 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39f6b39c132c4b55a4b9dbdf1ee0df7c : NODE_FAILURE Node request 200-0000038469 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38ea14a17cd44d58b61f5f958bcc4717 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/15c49828f99549f197de380c890f91c1 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"f61c6ed7b90a3bdea7cd249da929337ddf353345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 12:55:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c5dcce534d387d9eceb84c1c517fd100d9ba825","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 12:55:28.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1699/","accounts_in_message":[],"_revision_number":5},{"id":"6645d9fad60fa7d384efaf43c5b1770fcb35c124","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 12:55:31.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"7a8bbd305b939daa0fbdf3c2156da4b3fd987369","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-14 12:56:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"fa74b3e4b52025b6510b862fe3f916bebeac9ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 12:59:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b9dc8d009f86e05fcf5266bca6a5b5abe4981cf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 12:59:33.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1700/","accounts_in_message":[],"_revision_number":6},{"id":"5cc0c8b39ba65df326b7a8d2d534949307dbe3b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 12:59:36.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"86ef4219e7155361d867fd6615cbf6feefbcc6ea","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-14 13:00:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"8d462a61490923a81910431c973ae7e46a6b0955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 13:01:52.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"806dd13ada8684c35c44c6786b736d4b8cd89532","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 13:02:03.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1701/","accounts_in_message":[],"_revision_number":7},{"id":"974f49b923ac17f0ba5977bbc74bad083374a09b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-14 13:02:06.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"450adf1783137ecd2db23874e47a7befe6937a58","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-14 13:03:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"9fba5f7236fdd6682b4422628db40019dc48d635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 13:20:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ace16b21237423ca26a5986f831a8e9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd1cc83655614a18b7a415ffe61d96fb : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a4b5693dc8c409c9c075215410bf7a9 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dcb3dbb4ce7a43aa855dc5c7773c8981 : SUCCESS in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"166de6eb15bf4fa9a68f1d2da4a762e5a2aa724d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-14 13:54:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/928846 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":7},{"id":"5095d2a3aceeede0dc4e96c6f58c9b18ef8a29b7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-14 14:07:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":7},{"id":"678de50e72e7d5f61c003fee45a1b2427c1ed47b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-14 14:49:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e97bff98ce884530bd7a44d24edecf44\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b14b837215594de4b317e885155cae70 : SUCCESS in 41m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9acb7ae0d4d240f8bb786720737942b2 : SUCCESS in 51m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65c451ef711048a98e914abc4e6a215b : FAILURE in 1h 43m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33a5b01512e340a4b583fd98c7844c03 : SUCCESS in 42m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7217b4e4d5484083a814040d24dd0067 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7034f1dd612042ab9d24a771109d2e1d : NODE_FAILURE Node request 200-0000038536 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b4e4e00a4d34f5aaabd55750870bd68 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02c1098c5d4f4efb8f09090d9c05d0e6 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"4aa2f6fe090b55cc974e8bdb8d3854eaaec844fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:26:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01c07a224b094add97e31fa4bf895ddd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e796476e867348599750ed6d601def96 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985af56ba85346bea77e83446de09ad5 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d184a7d4505a49f4a56d8d02581fbdd3 : SUCCESS in 7m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03f92c3d2c074cc691488b53151ff899 : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff915a0fcd5940fa93babd707a2ff50b : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628494927d6046b39fa65e67c3b1c77e : SUCCESS in 26m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a4094ee409c64898baa23fad9543056b : SUCCESS in 41m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d35abbe85cbf476eb918cd7cf6f3aa0d : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce10b9ef9a284948ad556da058426374 : SUCCESS in 1h 27m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d43c8464c894db48d9688eefe3b52ea : SUCCESS in 55m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beb1801596ae4879a71bf106fa1b1151 : SUCCESS in 19m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/155b538b58744723a04abd26b6793882 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a9fb35a709be4dbba136298995e0a8bc : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6893a22ee12148b681bd8e0216819d9c : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e042d8ed0f046b99e29570cc0461f86 : FAILURE in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/49680355a15646b7a59dfc979725f84a : FAILURE in 5m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a58a0ad9eab84484a752906060773c7a : FAILURE in 37m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f88c8bba9fd4e95963ab1a77086d2a3 : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/373b7a57482048c780670ec8b053c1f8 : SUCCESS in 1h 48m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/829840f7a1fb40ef86c0ac6ffd7df07d : SUCCESS in 2h 06m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72d93bb3e4e7444e92ac6438907744f3 : SUCCESS in 1h 28m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b9b5d90697a94593b38649ec06893db7 : SUCCESS in 1h 55m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9af321fce3194f88a7dc97870fa528ff : SUCCESS in 25m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a8dd932031884ab18b622a5399d7854c : SUCCESS in 32m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/341236f404874a3aada1894ae9bd73b4 : SUCCESS in 1h 25m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef6ae0fdfed2490d91c0eab5414777b2 : SUCCESS in 23m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/817556de7ceb41fd8b7204fa39d15fff : POST_FAILURE in 2h 21m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/88de428a26014a7a9930a96a69d504ef : SUCCESS in 1h 43m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43d769134c734beebfa32d5a84550359 : SUCCESS in 1h 41m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe1c5446a0d7454d8d27a2190186f96a : SUCCESS in 1h 28m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b78ccb0ed694f7aa4cdfd72b21f328a : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":7},{"id":"64d521fd6fe30760b65370e65a4ccfa5cda8c7c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-14 15:54:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/70a4801a2f334f3bbc4064ed30146075 : SUCCESS in 2h 48m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/545e9634fbc04d5e8428667486d219eb : SUCCESS in 2h 08m 05s","accounts_in_message":[],"_revision_number":7},{"id":"ec486507b016421cc2391697dbba324a0163b171","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-14 18:06:42.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-3/FC/ : FAILURE in 7h 5m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-3/iSCSI/ : UNSTABLE in 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"96bb43a3ae92ddf4670c6ca7eaa9a032a9c074dd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-14 19:44:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/7/20250415002121/NEC-V-ISCSI : SUCCESS in 4h 03m 05s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/7/20250415002121/NEC-V-FC : SUCCESS in 4h 22m 48s","accounts_in_message":[],"_revision_number":7},{"id":"916ee13f03b8c5bbbead93513403e567f328cb65","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-14 21:06:26.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-5/FC/ : FAILURE in 8h 10m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-5/iSCSI/ : UNSTABLE in 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"efb0726f5a581c80102ba673bb982559cf785091","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-14 23:19:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/7/20250415022614/NEC-ISCSI : SUCCESS in 5h 24m 59s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/7/20250415022614/NEC-FC : SUCCESS in 5h 53m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a0a7dedaf51e0e7b0b90a40238198385440f3f0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-15 06:04:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"b84f16307660c89fe59efb6f1bf0ce68c08b26b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-15 06:04:55.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1707/","accounts_in_message":[],"_revision_number":8},{"id":"2ccd7ac7eeb1ed820a26888046612c1e8ceb415d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-15 06:04:57.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"07a6ec951f8a2a67f774f61733fb3f97b83e5f69","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-15 06:05:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"f5d448863567f322d9abffcefd76c2bb1cac6ea1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 06:21:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220abf94cf864297b71e51ed4a5b8c44\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40cd976243ee482dbe47537967109c4c : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4b82a48d91e4465b6ee5662916714f8 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/632540a2f23e4c1ebcc3c0cdf9ac52c6 : SUCCESS in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"415ab8c0424dd66c6e688a41b3304d56835c827d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-15 06:40:28.000000000","message":"Topic set to bug-2107329","accounts_in_message":[],"_revision_number":8},{"id":"0e05849d89c3f61bbc0d9f76e6de7fa9b8f9b0af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-15 07:32:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/928846 : SUCCESS in 45m 26s","accounts_in_message":[],"_revision_number":8},{"id":"f3ab8c75f135ec570ff4fdd59d4c66d57d0a4c7e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-15 07:50:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1645b27570974912acc6007c6e584ce3\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1253d7e7043e45dbb5ac4530fe0042ab : SUCCESS in 46m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c634d58fbed49dc8503fba0d83178f3 : SUCCESS in 39m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1ab08479e9d4aceb51ac0febbc75309 : FAILURE in 1h 45m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd5e15fbc64447deb5edabb061180359 : SUCCESS in 41m 18s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ae9095005d84fc9b7137b218881996b : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c997d543e75841e3a6ca685936de4456 : NODE_FAILURE Node request 200-0000038655 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eeac0b6106784dbab61205dad55dd29c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc8cf917a9ff4bcfb65ccbcb8d7140c6 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"d27dca736ab465b07ef212785935c1f07747ba72","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-15 08:00:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":8},{"id":"61c09399fb70924320f1ba25d49071d3eed786c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 08:33:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1322cc548c5443d0af7d1d0cda7202f3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ddea869736024e6a9ba018e41fb2c2ad : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de12c375a594e4da254884c32d2a7ef : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7f776feef704459bfd3191f7ff2d74d : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32c050e7cc1f4f509b4fc81d21af536a : SUCCESS in 14m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50a7ba24e24d4cbdb5537bb8e81d4d11 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab69c28dafbe456283d315ba31c08052 : SUCCESS in 6m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/113b88101147431585f18bab70590791 : SUCCESS in 55m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc47ad1bbf4c44cb82f2830848acee08 : SUCCESS in 26m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/902b417dc64d4bde8cf0fd73d3f49e9b : SUCCESS in 1h 27m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb4c9605e97e440abf2b7644fab5171b : SUCCESS in 32m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe92e68fbf4c4e26a52cf416acf0f871 : SUCCESS in 17m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03e949c42486408aaa99e94c42a04289 : SUCCESS in 28m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/368c69d72c6349ab8538f3b5a9bb6e86 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4faa05c40f424259b95d228f2fb69f41 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/241b199395c949f48c218606bbd06bad : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddb75e760d3a49baa55d5cc0a8c5f2c9 : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5db6769d5c904bfba307cc822bb6b6c6 : FAILURE in 30m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8c9c14bca914a5cbc6269191a367bf7 : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2faa56e0872043dfab6f9c8a1a669f42 : SUCCESS in 1h 46m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3cea787b3004ac3a9a8247b24bf5534 : SUCCESS in 2h 23m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b3fe4fcae4bb41848c79b4606bdd878b : SUCCESS in 1h 41m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/204487acd8954c5ead70557d6895d879 : SUCCESS in 1h 57m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/749d49e5f2a848ff8ae58a0d6ff57569 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ec4b94c2f63e479a8f1ef155fdc9a779 : SUCCESS in 36m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c022a68cdde64e0a92b780ba1e1e0d89 : SUCCESS in 1h 38m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76d3db1feac84dd99a2eaed349fc1b58 : SUCCESS in 40m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/987a3ac57d1b453789158398cff592f2 : SUCCESS in 1h 08m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bf832383b282413eae77075ab69115bb : SUCCESS in 1h 11m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a1cfd1e4d2245cf8ce67c07cee333a6 : SUCCESS in 1h 43m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/220d527fb41e42de82b681b23448dcf4 : SUCCESS in 1h 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1131d66edfee47768d181e0693e371e0 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":8},{"id":"1d2c8caa002a09841a5b88e643571dc0ab3ed96d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-15 08:37:41.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e266cc6bc8ed49379ba934c78bb8e6aa : SUCCESS in 1h 13m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0cbb1812e4654f0baea1fbbb2e287812 : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":8},{"id":"7bf6c95674282ae90be3732f5b4c14cf3d0be84f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-15 10:34:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"31540b0ec3ce664a5f4023ebf3af67ac49c9e810","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-15 17:47:35.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/8/20250415224324/NEC-V-ISCSI : SUCCESS in 3h 59m 52s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/8/20250415224324/NEC-V-FC : SUCCESS in 4h 04m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7487c7840f8bffb8eecdc5587f49f77bb4f28c45","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-16 07:09:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/8/20250416101408/NEC-ISCSI : SUCCESS in 5h 28m 54s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/8/20250416101408/NEC-FC : SUCCESS in 5h 54m 53s","accounts_in_message":[],"_revision_number":8},{"id":"855590940be3a7455f887dd5c8f1fe6adbd75e18","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-18 09:18:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/928846-8/HPMSA-iSCSI/ : SUCCESS in 1:58:13\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/928846-8/HPMSA-FC/ : SUCCESS in 2:08:01","accounts_in_message":[],"_revision_number":8},{"id":"c301f3f9b90c893cf334468341b18d344b679ad8","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-18 13:45:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/928846-8/Lenovo-iSCSI/ : SUCCESS in 1:58:42\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/928846-8/Lenovo-FC/ : SUCCESS in 2:07:00","accounts_in_message":[],"_revision_number":8},{"id":"78dead7509c78440eb8e3cef2a871846a3730b69","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-18 18:11:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/928846-8/PVME-iSCSI/ : SUCCESS in 1:58:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/928846-8/PVME-FC/ : SUCCESS in 2:06:38","accounts_in_message":[],"_revision_number":8},{"id":"f22e835da25794618727dc9a3208138f1e371a72","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-22 13:20:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a36a73a4c7619469064f53e5130c666ae9b10cbe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-22 13:34:33.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"681680073af72cb04b177535059f33c33e29502e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-22 19:05:15.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"48f660cf90bdc635aaa32fe157d14729fc43d8e9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-22 19:10:45.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"a4e223bbb030c304cc1c70a7590817fc425e0878","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-22 19:26:14.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"206c69e4872a06e9a3e23f99910425efe2f3c325","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-23 04:06:28.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"05855800b430fe8c3e90fc9166df869776c7f67e","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-07 15:47:57.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"742a9f7916b8afa049ac3d0c8fb7afa1b453d9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-29 07:00:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"55904ed1d0277f6c8642add447f9f02ddf287e7c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 07:01:08.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/351/","accounts_in_message":[],"_revision_number":9},{"id":"09f1ba1e1b0180dcfb5777bb0a82fc173e90f6d7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-29 07:01:20.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b416416cf039b94e1d5cb2f4294e7666aa64828d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-29 07:02:16.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"51dab55df040d84e85a87d7a4de9e2d3bcae2910","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 07:45:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/582b8e56741b47009d7a101dd9d63319\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c56d62d4722544f7bd5fe56402ffd7d7 : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24bcc5f35fcf40a99b6b9ff0a42e05c7 : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e77afedec404028afce89f25abbbd7a : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bb9b46c1d4a8a7bca56925e612f8f4fe40aeb495","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 07:56:54.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/928846/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"5f8469b9774ab2fb492f69975732577cdcb2de47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 08:01:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1610/.","accounts_in_message":[],"_revision_number":9},{"id":"88d38dd41211a1f32aa1142e8825a4cab94a9e32","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 08:04:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1744/.","accounts_in_message":[],"_revision_number":9},{"id":"68579aa05d260d6a44f8989732aec3a1feb1be5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-29 08:48:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 1h 25m 59s","accounts_in_message":[],"_revision_number":9},{"id":"32224bd13fa9e9c39ae934cf50cb49fd32090720","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-29 08:57:38.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"271de4116285f764cbbbc869eb9260cd40cd8fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:01:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ae9899678a4ad8a481e9a84feeb51e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/81e09a7731ed4b78a402f72599a8ff2b : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873c3419379843aba1ced0ce0889175c : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85ee5e2e5fa84223a86093e917f7a78d : SUCCESS in 7m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d95e9042ead40e2a24ec4d69036e07d : SUCCESS in 22m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2605fba77dd94d51858fb2ffa13fa72d : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6be22c140e44b29018d455db104f4b : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ddd39702287c4a5aa32e39e86ca306ec : SUCCESS in 58m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b83d5cd6c6494049ac387fa3df4c7026 : SUCCESS in 46m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9aec01c990944c818713c131ef4e44dd : SUCCESS in 1h 52m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db92afe5ea8f4b3aaef6574f90adb985 : SUCCESS in 58m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61470e9016384f9088915138b79582eb : SUCCESS in 14m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bbcaf5be6cca4cbfbe974136c792a19e : SUCCESS in 26m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6779447eaf64459c9be637f6fd0f4957 : SUCCESS in 8m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a1c65857b1a403bbedd0838a744a83a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c23653a703a4206a8e124248216c8a8 : SUCCESS in 4m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ecde6a27d60f4cb7bce5020525831009 : SUCCESS in 11m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4b732ff5c1c74f3395829924aa54534f : FAILURE in 31m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f64175cb1ddf4ffa86852f6eaec46f52 : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c73cdd63fa1f4539ad102505377d7956 : FAILURE in 1h 01m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48fed491ebd944bc8fd1af79f0f5fc04 : FAILURE in 1h 38m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef8ac85fbaa64baa9fd9675615a10bcc : SUCCESS in 1h 57m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2b7641b478734b4c8a5c46d71c489726 : FAILURE in 11m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4d51dc7c4920431684c3683e85dbd249 : SUCCESS in 24m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/adf5819f9ff44f0d879fae1bc88f6d3c : SUCCESS in 1h 09m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e698759f3574e2db6af67013dc78494 : SUCCESS in 1h 42m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c0bce0982f1f4670a203ee3ad08c61c0 : SUCCESS in 29m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd33d9248d9b45babc3a36bd809374b7 : SUCCESS in 1h 34m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8a6f546f299f4e489c11ced79a028a87 : FAILURE in 19m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec281894ffde45ceb340383e9324dc01 : SUCCESS in 1h 40m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7365a2906aed4154800c6a59844c2bcc : SUCCESS in 1h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/310e61bf5cc642c4bf72a8785bf2c5ee : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":9},{"id":"35a0dddd66a91d9f9736c94e9c01428a3a3dfddd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-29 14:08:17.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-9/FC/ : FAILURE in 7h 4m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-928846-9/iSCSI/ : UNSTABLE in 2h 8m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"97511d78fc26a3a9237d914795ebd8621123d50e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-29 16:04:43.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/97b2ce7c0d2043739a87eba6440e70a4 : SUCCESS in 1h 16m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/56c28a0de6c941b8a432459adee7baab : FAILURE in 55m 23s","accounts_in_message":[],"_revision_number":9},{"id":"15670625106a620d6def1d7b661166b9d02a3ca9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-01 14:19:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"63913c3978a21298c1ceea3a6491754d0cf6264e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:46:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb9a644e1f14b2d8ea98dc27706e0cb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23146db4d006466e9724b0068f38f75e : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/54308b7253984fccbbaee06e47078bcd : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d45c03780f53459aa768bf8e2bcbd46b : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d4bb0d3ab9ade9a0b8bb44c414cb3c7db1d8e2c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-01 15:58:00.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a742193b63bf465f8bcd2aba9d398016 : SUCCESS in 44m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bc39bdb4dcfa4f5eb2789e722caefc45 : FAILURE in 1h 00m 48s","accounts_in_message":[],"_revision_number":9},{"id":"29fd7de7a11ab5be47e972929d53428db6344671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 17:09:06.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d126fdc92eaa4d9b9fb0422489ba10a2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0d1f9ce25f154870b6e5fdab7223c89a : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be36883de5245d1bfb2f1292c7d4c92 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e50f9a4f12a49a98c4a6b65cbac8050 : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6e7c4483058444f8b466d7293ddc3c9 : SUCCESS in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bd725d42b484114b9073b40ed4bbf3c : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac3bab75e87474dbd8ff0ddb3124b22 : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2c332a8c53b1475fb73802d5474330d9 : SUCCESS in 57m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66d858c74d8144ffabf0aba3cc766aab : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc32561007f04bd6bdea565d1262e377 : SUCCESS in 1h 24m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d6e40a8165a4929869f037be0539d0b : SUCCESS in 55m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7909ab46e1d341e28c0e63500d8026d3 : SUCCESS in 15m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/716a5b18f02a40ceaaefc87603ad1cdd : SUCCESS in 23m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/74dcc7e4ba07489c80bdde472c77681c : SUCCESS in 6m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d78a91b83a2b49438011830aed9c3328 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/059c41f80335428ab2cf35044d30810f : SUCCESS in 4m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8a96a9df9f0b4a9a83ee444c09f3b9ee : SUCCESS in 10m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b2073ffc1d94d37911fb3bb06a2b2f7 : FAILURE in 37m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d171cd7da5384391868a960fa93ea10c : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f0a4c20cac44cde99c72ccf5d3e0da5 : SUCCESS in 1h 17m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de277428a361420487d4d251733442dc : FAILURE in 2h 43m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/259ef0e81162437eb46eafbf66fd9c6b : SUCCESS in 1h 24m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b4ce719e58240e1ac23da5ee427b759 : FAILURE in 18m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/311f07cb0c4a4b179d3804b28a6de696 : SUCCESS in 29m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d97819eda984e4a9b0dd64786c31006 : SUCCESS in 33m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aa6ed9df64fd4f75af542e1302185859 : SUCCESS in 1h 32m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8d09c322f3804b1e9845260775ee8d0c : SUCCESS in 43m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d196fe0c5e3e4982bca563878a49f7fc : SUCCESS in 1h 33m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a20d354a4f7e47baa1f658aa30a34166 : FAILURE in 25m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79d17947d85a457b985b6ec7093ec9fb : SUCCESS in 49m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a96e6f47371c4a579e44962d16b9fe52 : FAILURE in 1h 23m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2346de6aef6e42d8b71f168745bba8d8 : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":9},{"id":"86f8db5f95e28578da21601f67d1a984d3da002d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-03 06:40:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c32c9648d31e99f0d8ce9a8f27a56d0b3a5ab5d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 07:18:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47522135d6f54f93839e5317c7794878\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f83498dadbe14f26ba8959056edcda01 : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27d286569db84d2b9d32fa98747a4921 : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61fa6fa3a8c94a6d9af7b7bf6fc0fb89 : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9a0cff68d02210f4342b14067ada9e6d929d98bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 09:09:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9960b732adc74ec782718206f22a7ac0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/922d19f548cd4cdbaf295ef7658e21a9 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9519794258d741d097eae4f980f77c8d : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0e693ce3f9e44399ecb2fc8999e10c5 : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3db00d992684476ca5e3bd9fb7e42d0e : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/547f3c9215324ecab5d2b2cdfe10180b : SUCCESS in 14m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73fb099cae9b488f8a2cd9ca7775e2a0 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/adb3f2bf65954d72b54fadd8da791f51 : SUCCESS in 56m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf124b66d1d345bfac0ba446e56ad566 : SUCCESS in 1h 06m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bc8533efb8c4e3ca4b4d3f6380b7f89 : SUCCESS in 1h 39m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4df1c9dddcb4a6bb07b51ae13d4503e : SUCCESS in 37m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f36b809774bd4a028d48897d5e573129 : SUCCESS in 15m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a821d8883b904384958b6acf051d62dd : SUCCESS in 16m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f1a6408d56e4fc889f272dc9add0de8 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e0b753472ab4c0eb487ff42fb74aa9c : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/926c8adf18db48e29088264718f5dbf1 : SUCCESS in 4m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a533907cf87647e39e413c10e74bb0f1 : SUCCESS in 8m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/72a96ad57b52442e8f4b9a1b5b42986e : FAILURE in 37m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93ffde92a0ff4066b328d62b406048d2 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cf3d8b19907e4f09964542c2d52deb4b : SUCCESS in 1h 08m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d02f37d510d47eabe7b861c71a902e3 : FAILURE in 2h 26m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94bc709854ca46ddaec274400a833819 : SUCCESS in 1h 46m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eea732c51a2948129aa34ba811f82b3c : FAILURE in 19m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d605d25300b4400fab244918c95be4a8 : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/61fd81b463f34afea5a71f22d72f3e64 : SUCCESS in 1h 07m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/84bb5eba663d43238f220861bf6e5b19 : SUCCESS in 1h 48m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9aa402d7f884ed89236db2f9684cb30 : SUCCESS in 40m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25050242b2d0415e898c06ad4ddcd51a : SUCCESS in 50m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1bee73ff4d33428585d36fea6862120d : FAILURE in 14m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/18c77ecd6cf54f978ad058da768f2c37 : SUCCESS in 1h 48m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7c569b5f2044d968f6359841648434e : SUCCESS in 1h 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d828b362a5f40f491141a80c9e57797 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":9},{"id":"6ebda17fea54c677f2c6ebc055cb7b757eaadb86","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 12:23:59.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/33f1f0757b104978a4f9596e6584b1d7 : SUCCESS in 47m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/333a560ac12a40069b80756bf007b552 : FAILURE in 1h 04m 32s","accounts_in_message":[],"_revision_number":9},{"id":"0548ac64bedd1a42a1e042c7d452319277c1477c","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-23 00:09:44.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a3500557beeffba0e461676478cdb7ed3073180f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-23 07:59:29.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"deb6f59e3ca8782b404c78893537b48826de26db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-23 08:00:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"a8488de000d067edda488b9c0123a9ed2c94cd34","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-23 08:07:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b579a963391d04d4c16ad29b28dccde635f8574c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 08:43:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c1dab043ae46538ef382c94995a8c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de1586427e164ac0bbabc74792bb5f3c : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9faf39a0cfa64a60bc43e6ee353bd065 : SUCCESS in 32m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d3b6cb351a741178b3bb5324aa892ea : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5a3abdb316cf43e5758d493d30fa53c88229dde6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 08:59:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2111/.","accounts_in_message":[],"_revision_number":10},{"id":"220f46a50b440a6bce188d9ea380b9050c56973b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-23 09:00:45.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":10},{"id":"8ff4c91390724fbf0992da27c7f110a6b794e026","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-23 09:24:46.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f9f4f8bba50c4dc191611645f14c226c : RETRY_LIMIT in 13m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/65abe16c8f1b451284c43d8f0c76ca12 : RETRY_LIMIT in 13m 09s","accounts_in_message":[],"_revision_number":10},{"id":"0e096aebe323ab6bdc61a26551d7a70003294afe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 10:09:43.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2225/.","accounts_in_message":[],"_revision_number":10},{"id":"969ea412ab9c98064f6a31476e8bc96eb9541ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 10:39:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7457cef69af747cd930ed19c627d22ca\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2ed602ebf8924b55aa5a7ca9a5e4b9ae : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fdab36111247f9ada6d21e4066674b : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b3a70e190db4b3fabbbb409fb1ac2bd : TIMED_OUT in 40m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/642036bda95a4a7296f95c6e7302f317 : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94dbb8bcd4147baa615ed2696edf730 : SUCCESS in 15m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/b17065ac2f6c4b99adfb71536e0f355c : SUCCESS in 59m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9746691da26a4956be4e42a5cb6b8467 : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45421ca57b734d2fbf48b91e1a720547 : SUCCESS in 1h 28m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e63837831d8a43728b2d10234519bf31 : SUCCESS in 36m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cce326d67cf485aa319c1ab20270374 : SUCCESS in 12m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a896899f5014aa2b880b76faf90daaa : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ff265a7dd394d4e9304876e10b96f06 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97301e3b897b490bad00e23cd223f16e : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09a0d06bc83340daa474ca974c8bc3e8 : SUCCESS in 6m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/56dc1800f42040969668f54257aac3e1 : SUCCESS in 8m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7de5e265680147da94e0ef4055600e6f : FAILURE in 38m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89e7f2c1b95a46118af48acb1b8eee95 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e94e85dea5664fab95f7f8192d699708 : SUCCESS in 1h 45m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a6b18d720c4d4f719204fe62a382f223 : FAILURE in 2h 35m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f281d83782a64f8d8358122f30907bf9 : SUCCESS in 1h 54m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f3601369282041fbadf1ed35db58b94e : SUCCESS in 1h 55m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5fc78d77503b454e83ae84f5e3c67948 : SUCCESS in 36m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9895e41c70ce4f308add799f7930976b : SUCCESS in 1h 00m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/081af84f5bca46e9b9edf752c9185c05 : SUCCESS in 1h 51m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6f50354a78314df3b2ba38dab6cf5073 : SUCCESS in 41m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba2b69b3c5ee44bcb4313f6f96272b06 : POST_FAILURE in 2h 23m 13s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/93bf554447ed4d2dad2e8c57b3633d4e : SUCCESS in 1h 42m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c64a2cb23914aa4ae72380ea5e59887 : SUCCESS in 1h 44m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01dbee6d6c5e4627b4827877a6a22dca : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":10},{"id":"b6fd2cfd6aecffb17897b3967485a59aa62691a9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-23 10:51:22.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 1h 12m 05s","accounts_in_message":[],"_revision_number":10},{"id":"4bb3758c67d5d56aa7d14619c95408448d5cd4af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-23 19:02:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2db234ff338213ccc03aa4e2ce33c8ef1d3530bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 19:39:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e124fce2bd4444c48a039ffdc541d836\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/392dd1c44d5a4b279210ce870d6d69c9 : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0ecd0e8e3d9407fb554bf83a7ad279f : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ffff16e4d7e841c9baf38927428b6a38 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fecb6a308ea4a0ec8cde20a4ddd543fae2f561e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-23 20:36:31.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1c8838d519184714bc14279c7ffd3406 : RETRY_LIMIT in 17m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bc31a98065cb4b649ce08748b05c939a : RETRY_LIMIT in 19m 19s","accounts_in_message":[],"_revision_number":10},{"id":"985b2d2f97577e4d9c4b321d54a51db47a8f6f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 21:56:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957188374dc044eba4ee6a700f4c6ac3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c54e808f2d44773a6299b3a5bc031bb : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ad0c9a2ab54256b736c5b901e30b1e : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b6c168ef1be4d2ca1fbe4af97cf49ad : SUCCESS in 10m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63dfc64b71cf4c3f943040f706c977ae : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dffe0c73a26407e8724608cf749a75c : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a9138b182f93459482b1c947e72d7930 : SUCCESS in 34m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31523f130e03430285a9c1cb670ee74f : SUCCESS in 35m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1eeda42f51624a46860cf08856489acb : SUCCESS in 1h 27m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a06db9d0d874ef5a0448fe7416513ad : SUCCESS in 43m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/680b9090b0c14685953568f46439eb78 : SUCCESS in 13m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4915142e9f094e19a0b34c2fa41e3f77 : SUCCESS in 16m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e49df30a210d4b50bc7923c154fdb548 : SUCCESS in 6m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be87c94057fb41eaac4b3382f640fa47 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5c88310c29d48d28075425443dae754 : SUCCESS in 6m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ea99f9ec8a564bba8cb9eed0bd54b25a : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7cbe1c3470fb4c3e86a9f300a427151e : FAILURE in 36m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e42e4309d0624c4ea5449d5cd495b12b : SUCCESS in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e402dc20dbc14700bf7957697d6553b0 : SUCCESS in 1h 51m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4274fc25123648919a36e495a51c2f5c : FAILURE in 2h 23m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a1dffaca3e146a1a79421ebe3d364e6 : SUCCESS in 2h 00m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1333aec64b4e4438adb93a8eb3ef514f : SUCCESS in 2h 03m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b4a13e7605844829ed4542cd8b1940e : SUCCESS in 38m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3d251c2afe7442ba56f92dd2dfe40ae : SUCCESS in 1h 09m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c836c81402e64f93a7fb4516209c94e0 : FAILURE in 1h 41m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/077f260623e044de85a8bbd39f3ce35a : SUCCESS in 41m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa7f5e8556b9433ca630fcace914df1c : SUCCESS in 2h 10m 04s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/685fd935c56b4a9a8eccea925844bfd3 : POST_FAILURE in 2h 49m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/756589d0656747d8aa2e27d13a9b3af2 : SUCCESS in 1h 52m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34053c90a0fb405c9fb571823fb29702 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":10},{"id":"d2e6f2965d1d3452acda7f2b9df5a1f408927af0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-23 22:03:15.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_10_2025_21_56_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"e0deb06a8d197c95d098ec036143c8e684f30fb5","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-24 05:54:53.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"97188b216cdfac391a740fa26a79da48e48d7583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-03 15:51:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"110d79c8307a83d6f51d6490a8390da210b2e128","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 15:52:06.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"e1d02f890a33be1bbe3025353075b087fd8ba07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:25:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df097bf901a422ca3a9793903a33446\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67e29fea6d814cf7bdc3f52b2a07a0f8 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99a7b663f37a470fbe8c5cde2cbbe373 : SUCCESS in 27m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9597f826e444216992a3179277a8708 : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1d92421d1a59193ba431310719cad0390044dabb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-03 16:45:42.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b541b23e8d154122a4a1c5d4a5e2215b : SUCCESS in 50m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/11825f08077647618010e4f2b110e161 : FAILURE in 50m 48s","accounts_in_message":[],"_revision_number":11},{"id":"5ebec33483afb40c3f6cd9d505cf76c0c0a11f8a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 16:50:15.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":11},{"id":"07b613ae9ada6e48d4fc06081a643a2c65b4210d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-03 17:03:22.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 1h 10m 18s","accounts_in_message":[],"_revision_number":11},{"id":"74cec848792a3346c69d67dc742d85400be5905a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:33:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a2a0a0e69a4ce6911cb8e04e1201e5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e0f88491ada241bebd8a942151407dfa : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae2c30787244a4e957ccd3584913b1d : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbba2ae1c26d4107bfcf92af2fd46dd4 : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bb2ebe5e7ef41fc8f24a39f5101fa96 : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd505d4af8f4277b87c03929fa8c6e0 : SUCCESS in 15m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/55f897dfad4b4672b9a4183cd917b094 : SUCCESS in 58m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb23e226a6074c80a8f48b69cbdec8cc : SUCCESS in 47m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/681b9af924d444ac9503aef92c0772a0 : SUCCESS in 1h 28m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94f2ecfd366a4f5886c05d248feb09ad : SUCCESS in 32m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e8c6491f3b41cf983216ddef41ccd1 : SUCCESS in 14m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b19b1c5e018d4a3c92891e4827c07d97 : SUCCESS in 24m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fdbf7374895c4afca416dd903a21def8 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee6340a17a5640d791d085b6bbe95172 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c241623950144b3ca4ab546b7c80d3b7 : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e23348585fe74d8985c78b66833075f0 : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08f73c19914c4c49881ce7b926109ede : FAILURE in 40m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/513ffa354c744697bdbeb655cf7549a3 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c4c1399fa75c43e2a837e1749da851a6 : SUCCESS in 1h 50m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d9d6365a49ef413ca1ddd6f64bf8fb75 : FAILURE in 2h 37m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6497318e7e44419c983ec2220016cc40 : SUCCESS in 2h 05m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5e9dbbf2f3854a0c818815c0675419e6 : FAILURE in 2h 23m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b020976d12f34e4a9b739b8d0b866657 : SUCCESS in 34m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89730a7f55d949ec87479509a7de7101 : SUCCESS in 36m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c9fceca0f294a1caebb100b3600f823 : SUCCESS in 1h 37m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48858b5890e84b0ebd062bac00c1b977 : SUCCESS in 38m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b8d15087f7a54372a3cb4dcca391e8c3 : SUCCESS in 2h 07m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ccaae001ce0345239c77a0ffef363c7b : SUCCESS in 2h 14m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcbbf9579e444b8d98b82fac1f7dea8c : SUCCESS in 1h 55m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f2c8b68ce72442b93f912a9b4425e2e : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":11},{"id":"eaf710d56f68c7b3c358fb81c85f742a9ca4f904","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-03 18:44:24.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_11_2025_18_37_53\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"9f830ee0491ba4539e64481274c54e6f235f8567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-16 15:26:18.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"c91a91ead62a3d74d8c605f53e9ff5e2585c9eff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 15:45:20.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2753/.","accounts_in_message":[],"_revision_number":12},{"id":"02cff2e01e41f5c50efb1b7f1d6fba29e19611d9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 15:45:20.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2900/.","accounts_in_message":[],"_revision_number":12},{"id":"9cbcbc8c07970e9653e15cd69bbe1285cbe839a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 15:51:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e572ab0f644fb5a6cc28659fd9db09\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b391e0963b747158b17f68916d8efaa : SUCCESS in 17m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e6143b2898e4c25adb960e6f722fa89 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e105ab4e256141108c5830385747ab16 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"979f2bf16dfecb35866e6909714d125df32932d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-16 16:39:41.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":12},{"id":"d94ed888ef181042b2e9f304f2ecbe84609379e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 17:32:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aa77fe59cec4fb3a884e135ed5f3556\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d178cbfc536f4bcc87dc7cf8aa083826 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7169a2ea8c47349e2768e350753dc4 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06cb5ec659e94775a9ea709402964fa0 : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ceea5baddd84e3e81ae6f6dc7c9487b : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9eaa169cda841a79461acd263e25810 : SUCCESS in 15m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/6d577ee53d0a4818bc67b3b7f553802f : SUCCESS in 41m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3bca991c7dd4ab2be8c6809e8ab8b93 : SUCCESS in 39m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4524dc8719414917a717edd4399dffb6 : SUCCESS in 1h 33m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b96fb4e35534eff9bb35d6a966ca75a : SUCCESS in 49m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/075eda0c847b45a9af0e22a049d9f41f : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7f7e87b92ad408589bd5aff4f47d80a : FAILURE in 24m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/32403faabdec49a7a56798496db4e8a5 : SUCCESS in 4m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3dd90b318545474485f4dea27782807a : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/425115396b734518b1ca8b8c5e7d8e8f : SUCCESS in 3m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4c5e337ce6e0489890b76088510fd7bf : SUCCESS in 5m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e7fc516fcef4b85b59e9b8f93a853fa : FAILURE in 37m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2eeb34c4c5fc4768b47fc0ee01b70670 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/279a1c93d6504857b50ca292c2a80a0b : SUCCESS in 1h 14m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/06009bba378245ed89a2428ebc31c5f6 : FAILURE in 2h 04m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0acf29d90bf14f9ba7370a0b740c05dd : SUCCESS in 1h 46m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d1b59c58e3f44aa19d5e102f3e96974a : SUCCESS in 1h 43m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/80611549befc4edfbf500bf7e3f403d6 : SUCCESS in 32m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ddee7d13a5ae4e638482237c9d6440f3 : SUCCESS in 53m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5bc726e90d64a80aed3042d2b9fbdd3 : SUCCESS in 1h 22m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/272eb4df3a484e4b9f571cfef48786b5 : SUCCESS in 50m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9cc3b9452e8e475e86b4deaf8231ccb3 : FAILURE in 2h 02m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/da63f9a06b3e476d9c371de78c1903f4 : SUCCESS in 1h 05m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8959646c29794193b5e32e5de6a87145 : SUCCESS in 1h 05m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc03101e9c794429bd4ef0641c8bbbb0 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":12},{"id":"9b06f8c44ff9b659562fd6650019b476ccd9e794","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-16 19:01:00.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/277baca0d87c42bcaf05e77dbbb5ab64 : SUCCESS in 50m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3dd56312c0534b67a0c70617ac72ad82 : FAILURE in 48m 49s","accounts_in_message":[],"_revision_number":12},{"id":"38aea5ba6bf0e74aacd8df23e522c143480ba266","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-16 19:35:55.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 1h 19m 36s","accounts_in_message":[],"_revision_number":12},{"id":"195f7a89304b28f8ea693466da5dc6f196fa7eb7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-18 07:40:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"22a33cadf9a7a3547d2ef5f01b9e608922a107d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 08:05:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991ac8872eed40adb7cb5b759df8be6f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94f7b08d11d44b7aad6b92f643f70f93 : SUCCESS in 22m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dee60b93dec4936bdf6e21e159503cd : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d028b8981d24d6abedeed3cc34515be : SUCCESS in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aaa0b0f2999f22a8c0f8708fb30a1f22061e3f62","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-18 08:34:08.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e23f2e74ba9a4e7d9c73f72acd9ce27c : FAILURE in 50m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5a8db77226a042af9436f56b63c6992c : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":12},{"id":"1bf2ebaa7288c29f37b320b4db65e13d1801e745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 10:17:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7e77edbcfaf49f3827c6f86508c3bf5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e6476ab334543f3a8ab54324a17f216 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4581e5bb96c4bbd842205f2b364b995 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29b3da1587754e4f80e188c0ba2cf22a : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0b186b3b8ae4ae7a59aa48232e20192 : SUCCESS in 14m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909be263faa64495a69808c8c35cd202 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0c063564880d4acd9f67b0ac285e4184 : SUCCESS in 56m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af4e8339a21047fda56481b14f90796b : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09f24d350a124e3c99acd7da95b1f1ac : SUCCESS in 1h 52m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb9b80db72724a18a95e2dc32331e53f : SUCCESS in 1h 06m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81546912361e4850b1a4f4ab08bf0a27 : SUCCESS in 15m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f8b5c7c77c64dcf93f15532a14bd5d7 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3251a6caa71b4e0f80ac9a0995e231a8 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf626ece8aa74510b6dde30c8c5fe525 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5462040a516d4bd79dda2eb7c239e126 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b2c010d790f48b19002eeae8f82905b : SUCCESS in 10m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/deb4b08fdb954e4799691c7c2485fa7f : FAILURE in 38m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/134f6a22c51b496dab23dc91ebd1f59d : SUCCESS in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77295677125e41c0b7897fec265772c0 : SUCCESS in 1h 49m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81d4446eb6de46caa692311ceffba722 : FAILURE in 2h 31m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2259a258aa049ebb427a1224bae95a8 : SUCCESS in 1h 36m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d4d86ced0321430db1aa6c5606c3a49b : SUCCESS in 1h 46m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/98b0526bc86a4f3fac1744ce4b284dcd : SUCCESS in 27m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/725ec8ef6ff646f5970bd8d56a4a49c0 : SUCCESS in 1h 09m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0c11c8d79b94dff98144910babe09a1 : SUCCESS in 1h 50m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/196f483300bc438b854c67bfde4de34b : SUCCESS in 41m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/56e2a6c1612a4e2299c9df145e194953 : SUCCESS in 1h 29m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0769319801f34a4d9cb91543340a98cd : SUCCESS in 1h 51m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cabaac7b6b454871b8e4dc857c957962 : SUCCESS in 1h 23m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/754d0b794dfa479db9abaa3fc5c08bdc : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":12},{"id":"3e7a3e79cd5dda42e727ec11139dc378160d55a8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-18 10:21:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-928846-12-fc : SUCCESS in 58m 54s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-928846-12-iscsi : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":12},{"id":"fb5b78fe3b99eb41640a8fcd84a8a3a126a80dd6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-18 10:24:44.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_12_2025_10_18_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"2038556a2b62f979fadc43740f6f38e4fbbd92e7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-18 18:47:32.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3a7f1a1431c872fda2c8b5e59f9bb61032d5ed8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-18 19:53:02.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"1630ce4b168dc08f805a627fa284e41bf7fc5a2b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-18 19:53:06.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"edfa00900ffb89bf9a8e797674e4b6125a804d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-18 19:53:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"34cb98979f9c9b11a92ca38fe5c755cccfad5c42","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-18 20:02:58.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b830dfac768246e48238926848c020d5 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9e7085311a4543f8b33c8d293d0302b3 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":14},{"id":"95a0d0fba3aef6403e871001de03846347854f82","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-18 20:20:28.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":13},{"id":"e3e3e229dcab44e4368da933093cfeeafcc39635","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-18 20:20:44.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":14},{"id":"3c62fb75909fe5594e55da11781f5b9d1137d33b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:49:52.000000000","message":"Patch Set 14:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/747/ : ABORTED","accounts_in_message":[],"_revision_number":14},{"id":"1c085b455b42619b67c51072c446a38c84329634","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:50:27.000000000","message":"Patch Set 13:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/746/ : ABORTED","accounts_in_message":[],"_revision_number":13},{"id":"99ad22d78fd472f44b7b16dbf20899b14546c101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 22:13:23.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b44bb79eeae243d296ed2dc2adee40b8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7cfb24c6f4ae4d59bbee20e15ac49eaa : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f9e58a735e4302bc4f3ecaf4aa3dab : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/951e7975c93c4332bf2dd2c749f9be8e : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77632a5478e9445ea2b147b40685d61c : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441879ebad2a4a639f4212438c998c0c : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/5883c6da40b24f1390d5975bb2d080de : SUCCESS in 32m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a80767321c274f3690951b2e7d289a9d : SUCCESS in 1h 09m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9365e9e92294a30a82767418eab342f : SUCCESS in 1h 31m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9906ff230b714125bca669586bc059c3 : SUCCESS in 34m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ec90a05ae34f9dbc4fe0fc13877177 : SUCCESS in 15m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d41b0d17eef3475793750ac361ebfeaa : FAILURE in 21m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b323e8b4b9240748cf01d7e3f567dd1 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba06b5a6f06d43dab893ec6219e45c1f : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef327bfba4c0476e9f754a751cc5ffc8 : SUCCESS in 4m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8437e017f6f44719b583302f925eda1c : SUCCESS in 4m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/16c3e011fbfc474498b61db59c9fd149 : FAILURE in 19m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48ccd6cf2fc445fe906a0831304b1446 : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a0c87a19f86491db3394fb0eb38916f : SUCCESS in 56m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/03b40cbdfe6f433e9aff0eec63f639a9 : FAILURE in 1h 16m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea98e9619904467ca50d7633046e873b : SUCCESS in 1h 53m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8a1f37a44bbf41bebb61227fee823b3b : SUCCESS in 1h 55m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/34d4f7fc05d944fa9ccc55d379f750e7 : SUCCESS in 25m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/61ac629d32b34270825f8a1a7e2450eb : SUCCESS in 1h 07m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/00dfa9bbae094e778ad41c4c173fb50e : SUCCESS in 1h 43m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d57cb2a832b4d8f8ce82dd05c20c8ec : SUCCESS in 44m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e498a63d69e84666b4ecb7e270ec43af : POST_FAILURE in 2h 17m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d32e30ca143a4338b24dc9600c725e93 : SUCCESS in 2h 06m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/319f49be9db14c11ba878f24ef7b682e : SUCCESS in 1h 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a52603fa9f47069b869711243a4d84 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":14},{"id":"c05c77dd706357d5c70a4958e6a8b13e1cabaad4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-18 22:18:47.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3341c25ff1764345941d7c877a1455b1 : FAILURE in 42m 51s","accounts_in_message":[],"_revision_number":14},{"id":"2754c9b30927c5bea8324582b21fdfc313142ddb","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-18 23:59:02.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver build/ba1d6ae89f9c481eada011becda80a4c : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":14},{"id":"2d9c303a7322496c16e4f0ea54406a94c4c54f4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 23:59:41.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c484128d09724682b4802b3bd2ff754f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10ffe01c06af4f978a81352697d7e557 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f4fe43c095b452a957ca751f1336423 : SUCCESS in 14m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f084fde5bc704d35b30ef42aafc66dd4 : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"58c41b349d247af44a3f67b3d5f132a88b465188","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 03:07:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 37m 21s","accounts_in_message":[],"_revision_number":14},{"id":"27d0e7f060dfa1caddb191ff2d383306c10e4618","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 03:18:54.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/169fffab28d44e83ba20f4dc706b7b04 : FAILURE in 38m 41s","accounts_in_message":[],"_revision_number":14},{"id":"6c312c5b4856d022a28a1b621f9650147ab2fd0b","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-02-19 08:55:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"23a3d82ff79907e317909dc075af440f1a98887d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 09:34:11.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":14},{"id":"da50d190e70a08a0a2bfd2a72ea2e5d1b9ed5cce","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-19 13:16:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2cef162bf6c6987b14194bc1466f10da8d0e27a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 13:21:24.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4417b5570b2047bcab541c4ca0f5081d : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3ac10fd11d6f4f799ae492fe097a7441 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":14},{"id":"7c46976ac4ca215ca6ca21153cfe20a0a81471b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:53:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba2cde569e884b37a0116af97660ff41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/540d5265768a4858adef59c8e0b71597 : POST_FAILURE in 12m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d20235aed924020bba726e5a56e87e3 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a540d4d7799840a4926c7f15a4e47e76 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e63bcaba34c2a98fb98486134c81882dd31f201a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:01:32.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12848f54e63345c1aeb0be248bebb8f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/72711bc7a63049bf9340fc5a18d8b17f : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183a2f7cd2d045769a042e415f888541 : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a5af58e2f4441d195df85b7111a83cc : POST_FAILURE in 13m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1aca6707d9854369b3220aaa303d081e : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c96eae9d30e4232a74ddb3451bad40e : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/972aaf041a954a468920c29a7e082a88 : POST_FAILURE in 1h 00m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdc487164d75426ba92bc5686bc3fc1a : POST_FAILURE in 33m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4fbc0eeeebd4a61bf14e61f814ae076 : SUCCESS in 1h 44m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2113cd167aad412c8d5c8eb18aae0c1b : SUCCESS in 58m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d04aff2fed664b5f86d7b75ff7c139f9 : SUCCESS in 16m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/85c576f7b4c5498ebad160fd51f93a8a : SUCCESS in 19m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7665ee790b364568b4a034a63048f2d7 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0896a7d4b78a4b5b88d2132522f5c494 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25ec8a94a82641ccbe269271d7b29db1 : POST_FAILURE in 4m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dab8bfa1cc094387b4e03cdf074cedf7 : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/20a460f8a16a41279de4b8d96ae0c3c6 : FAILURE in 21m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b6ee8ea4d4345daa87e59abc3de72cf : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e96eb11c5df142df8b823651b80e48e7 : SUCCESS in 1h 09m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fbdef2fa25f447a92fb2966a4bab8aa : FAILURE in 2h 40m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d8f9b609dca463880e088973b34b440 : SUCCESS in 1h 50m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9f88c996effe4431befb624f687696b8 : SUCCESS in 2h 00m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/54558de4997041e19f511a83fd5c8267 : SUCCESS in 19m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f90a6ad3824941c89451e5cfdb33280c : POST_FAILURE in 1h 10m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5c22a8e8fee43709765701b8e57889d : SUCCESS in 1h 41m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/edf6eabe3d6840998783cf97bc15c145 : SUCCESS in 30m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6d8288fb0314547a9bbd5a5d5b4d171 : SUCCESS in 1h 37m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d405fd582b1c40b2ba8da8964a1e5496 : FAILURE in 1h 50m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6e7bf7523aa4208bf3a992cb3bfd898 : SUCCESS in 1h 54m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8375dfca55474e248b4ebc3a9f091a50 : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":14},{"id":"29c4022113967e5be0b15f0b4db9cc306d401105","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-20 14:08:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f5970b3b23901bfb5c5c028213eb84846b3b9a6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-20 14:11:37.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"c8d91811b7b25bb0b8747fc288a932eb3ff3bbdc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 14:18:40.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ea4b0260b20240a6993c1907af42d221 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/50a03b30d7e24715855211a00d75ff7d : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":15},{"id":"dacf32e281716a4bfd184946d6f729258b408792","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 14:37:30.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca650cc59e714c62a5c672986643140d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4aa693a8432412cbb7cf77e2bf2cd7a : POST_FAILURE in 11m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84fc687135be4b34907d2223cd140be7 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe6369d5cf6e484d9d302ae37da435d7 : RETRY_LIMIT in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a6d5f47600ed6dd16f414fafac8c3c01d6d1f53a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-20 14:45:12.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver build/d5b4cb6b7f6f4407996398ea7c3c253b : FAILURE in 33m 09s","accounts_in_message":[],"_revision_number":15},{"id":"47200332cdb90a1a8b567bc2cf151ee0bf96a9f9","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-20 14:49:05.000000000","message":"Patch Set 15:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/928846 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"25779826bcd177ae9f11b0fd902069480eb8816c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-20 14:49:13.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":15},{"id":"248694bc75cabe8addd6d3f60def55f93bada3f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-20 14:52:11.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c9ae0a3926d641a392af1197817ce1b5 : FAILURE in 39m 56s","accounts_in_message":[],"_revision_number":15},{"id":"c674bd6685a494b2faba7173e02aaf1cf1415b7b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-20 14:52:50.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/928846 : FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":15},{"id":"7aa3abf58adbab5bed5aedc10485779917cd4eed","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-20 14:53:29.000000000","message":"Patch Set 15:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/928846 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"87f1667a8012f6a2a578803a71f165b1fda27e4d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-20 14:54:53.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3235d65b744444d0abf29ff4e916725f : FAILURE in 42m 50s","accounts_in_message":[],"_revision_number":15},{"id":"3ee658fdc2d36710049938e58817ec32a253ea07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:35:39.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb136366ba8441938681e432cdd02f41\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8016e1d8fa334780a25158b41e3af747 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed9a57ff47c420f83c582e41f3b008c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7c915d253a2493ab41f72be7016d37a : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cc2c2ea0c1d43e2bfbc124e41855af2 : SUCCESS in 11m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8bf9f381ece4f109c485d323d95f5c7 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de79a93618e84bacb3ccfc09df2f388e : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/e87861f0703e4dc1b691247198fac679 : SUCCESS in 1h 01m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15816738890646ccb5b419a6c94386fd : SUCCESS in 57m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e011b53db7f84a59947f5357eef4714b : SUCCESS in 1h 31m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b6ad70964ea42d393480e2efc71ae11 : SUCCESS in 1h 06m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce026fe8e2e4d0d8149727186204c49 : SUCCESS in 12m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c79d257c7e644dbe9927c6c9c72f0753 : SUCCESS in 21m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/05ef37bfa11f490fa1570da7121c5702 : SUCCESS in 3m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85e802e160e440a29d634c0cf07954e2 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b663f9933834c4d86d75f2e6ec111b9 : SUCCESS in 3m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b5ff68c6b8a842e491764ad5166349b3 : SUCCESS in 4m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a8bf2dbf5824b31b92599995831d996 : FAILURE in 19m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51d2e3d5dd2b46f0b413f9e7f37ea8af : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d2b01cef430423cb03a5bdf7b62fdd3 : SUCCESS in 1h 05m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a165f50f23d41f2bcc2a212ff366dfd : FAILURE in 2h 20m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4bb6a1865a9d4befb173eaee5ec9a8b1 : SUCCESS in 2h 11m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3a1dd87b7d6d45b094829324691ca2c4 : SUCCESS in 1h 48m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/af59e76441124b61b43eab39c1e27997 : SUCCESS in 36m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/588d078b24d84d6aa8a6b821b16f1df0 : SUCCESS in 1h 11m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab97706f7d364e64abc37393dad5e410 : SUCCESS in 1h 19m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ea94ef8776849c6b60922aeed180b93 : SUCCESS in 24m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/59b24ec0fce64c82b2ec686c2d3ae452 : SUCCESS in 1h 12m 22s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3e16506bc56f43da9c13c0ca39210d62 : SUCCESS in 1h 52m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7810a12e205a400d857bf7044d9483f8 : SUCCESS in 56m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3369e2f9abc6479f8321bad00c55c107 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":15},{"id":"8d463c4ed0e57f8afaef4b67ff8acfc090a1da27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-20 17:21:44.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_02_2026_17_14_45\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"39231e653de8f4e6e8c8175baf6614ca7c47c7cd","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-21 20:06:42.000000000","message":"Patch Set 15:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_794/index.html : SUCCESS in 97m","accounts_in_message":[],"_revision_number":15},{"id":"b5f211a61ea62ed512f107aa7eda8296d073a2be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-12 22:04:30.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"86531fbb6ab2963c32eda30c1b7cda46b7eb20b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-12 22:29:53.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e4ff9eb6e8014b21b222ddeccd1d74f5\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b51972716ed423f82bc3b2e0fac82a7 : FAILURE in 24m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22497143d44443c4a10e295d4ad194c8 : FAILURE in 20m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f4481e1efad4dacad4cf0a2a7765b5d : FAILURE in 24m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a4efb6e324bc448cb8f215891bbadfd1 : FAILURE in 21m 44s","accounts_in_message":[],"_revision_number":16},{"id":"bb8a0d2ee2983af50bcc2053e0095a61ff5c0547","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-12 22:41:34.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-fc-driver build/35fd6acf03254929a310c640d2f0cdb1 : SUCCESS in 36m 01s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/928846","accounts_in_message":[],"_revision_number":16},{"id":"b3d89370004abfec4a7f694325651fa74885b4f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-12 22:45:05.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/213e3fa404cb4e98a419b2cb5b8fe820 : SUCCESS in 40m 03s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/928846\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/928846","accounts_in_message":[],"_revision_number":16},{"id":"fb6d99f7b21f146c12227dd4020dada17ddf1aa1","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-12 22:53:42.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/e3fa617846f349b5990b6e16813993f0 : SUCCESS in 48m 24s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":16},{"id":"70e3be137a575bedce9466397c72e10531673bc0","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-12 22:56:59.000000000","message":"Patch Set 16:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/160 : SUCCESS in 52 min\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":16},{"id":"c5e3e2b4ce076c6d2c2e328cb79ee153a47e6b0b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-12 22:57:19.000000000","message":"Patch Set 16:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/423 : SUCCESS in 52 min\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":16},{"id":"be492b5ba2b2bdd5a37bd59a16a9c8e03633eff7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-12 23:00:01.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":16},{"id":"c389fefbf62f94f9cb671e033728b0e6308aaa8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-12 23:00:43.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1fe491bbfabe4a09b4fed35856cbdef2 : SUCCESS in 52m 17s\n- cinder-storpool-tempest-iscsi-multipath build/d3ef927da04f4a99b4b580b84a2a3159 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":16},{"id":"8df054ae91756c5d2c6d391c06c0aab10d0d4c35","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-12 23:12:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4467/.","accounts_in_message":[],"_revision_number":16},{"id":"6f70dc6149ea2e281e90281d5324e5897da2a085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 00:47:39.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82185d9fdc1b4965bb48a07f4ee18d8a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2849f69088e543648144ca03591e3e6a : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1af328cfe454b42aee5cbed1e011736 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/455542fd3018439ea5b9ee1dc6f61c80 : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42586497ddf64effba2cda693d04f121 : SUCCESS in 10m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82622e87eb9f498fad73ca31bc91addc : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241c8ce2c9df47c99297ba556c0f2332 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/673a4da1427149b883a166dbc12991d8 : SUCCESS in 1h 01m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93ed5751591843b4a53399aa1a6d1f2c : SUCCESS in 1h 08m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e81667bc4004807a0f39ef88947ef34 : SUCCESS in 1h 13m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1022656fc3ce4992a271b75b5a394724 : SUCCESS in 49m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65a8a5843961418c8f86ec715df32757 : SUCCESS in 7m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/259b3d9d097a4ef4b8dc7743d22e87e1 : TIMED_OUT in 42m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a3489f4ab704034af53857e3f115f5b : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f0cdc8f53454a77871b2dc587d0a712 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d54df2a4df1418d891f8170aff0641f : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8c51771cf6e7468e8e33954b676f3fc4 : SUCCESS in 4m 38s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/08b7a4d9aff44300ad995085ced2ac4b : SUCCESS in 8m 32s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7aab64c2d866438dae84b04ec00106a1 : SUCCESS in 22m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c27414cf89044ab3aed0a3f9edb3a4b0 : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fab348f9ca0427c85dcc12861af63ae : SUCCESS in 1h 42m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8c447c87eae547acbde9466b976f20d0 : FAILURE in 2h 31m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d507a0b73bf4425aa95289ec11553de2 : SUCCESS in 2h 06m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f7bf6aac0e7d44a9b79c8cba43daf687 : SUCCESS in 1h 43m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04e26da76d444097b2ad63d588dff1e1 : SUCCESS in 50m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcf02e14a7e54bb587963617b1c2b95e : SUCCESS in 1h 14m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc21b4e4665d4639aa66a05d07d879c2 : SUCCESS in 1h 42m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/33724b893ccc445c9a63c39a4c66ee72 : SUCCESS in 51m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3b48cb234c44708947b7ca8bd74f2f4 : SUCCESS in 55m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fce784c8e21406fb65c1f0599415b48 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":16},{"id":"24c40995448b54336cae90a0f0cff0bb7de40e39","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-13 00:58:32.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_06_2026_00_50_54\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"f73cce609c6d2e91f58e1a8f67c0ce0ca79b02d9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-13 10:34:13.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-928846-16-fc : FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-928846-16-iscsi : FAILURE in 1h 4m 4s","accounts_in_message":[],"_revision_number":16},{"id":"b7bde89bbdf4acbd53a2064e629c4470f6dfd030","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-15 14:35:04.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"396b0ceae28df31aba079cce5ed6f225f22495cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 23:34:39.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4576a88e017846abbc286c3ef881007f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e93b0a339e8145f488101802230bd973 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/400d633ab0804cacb36e1266d659af6f : POST_FAILURE in 14m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/650c336f6d384adf841e3d6d18e780d7 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"09c41fe3b0900c7adb9f1f9aa6dda8349c6f708c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-19 21:40:24.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/16/20260620004256/NEC-ISCSI : SUCCESS in 5h 28m 09s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/46/928846/16/20260620004256/NEC-FC : SUCCESS in 5h 57m 25s","accounts_in_message":[],"_revision_number":16},{"id":"86310b45f258149f75bb9cf659dc3b07f1935e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 07:20:03.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"4131815ccb946e5013316d8492dcd9ab268fe5c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 07:21:03.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"beff1f80203df5b354d340ccfc35f601b5bb88f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 07:39:13.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7218bdbd65a09c554e925b450b3ff0a05ad8ede9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 08:00:14.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f9aba204a64294b2e9d16fbe70fad6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2669be5be5874611b3ab060ef239e767 : FAILURE in 17m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c663ec00ed04df3962fcf85a75308b2 : FAILURE in 16m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a53482f2db2d41fc8893ac5b35bce377 : FAILURE in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1c365685fb799d2f5dc21c7e5f2135b0c3356f72","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 08:12:03.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":18},{"id":"f098bc67a1a84a48986feadd30046186563165f2","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-30 09:08:08.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- hpe-fc-driver build/4bde5b26eec640468db2fc262edbb5bc : SUCCESS in 35m 20s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/928846","accounts_in_message":[],"_revision_number":18},{"id":"a7456f84e4b854b40e84a0dc0cf5d1607e8bacaf","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 09:47:36.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/380 : FAILURE in 0h 39m 4s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":17},{"id":"51e9fd32a094a008042979fa9b72a5f7acd9d8bb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 09:49:46.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/286 : FAILURE in 0h 38m 51s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":17},{"id":"9cb5ac6781dcc784cf9f55f522385f56cc2a6633","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 10:32:23.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/652 : FAILURE in 0h 39m 23s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":17},{"id":"a0dd61ff61c52b527ac9c854e0e3ab0e6e17f065","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-30 10:39:59.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4576/.","accounts_in_message":[],"_revision_number":17},{"id":"c88ade4200dd13907e2232c26a7fa8e817697bb1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-30 11:11:02.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/a606ce11b7924879adb3ab76bcdc26a6 : SUCCESS in 52m 35s\n- cinder-storpool-tempest-iscsi-multipath build/52437cfeb33b4185ac0750dcab38128e : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":18},{"id":"073a1b1bb920c59a595d99f1bb9329179d9f16a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:15:59.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/807cf6689d974611b01c633082cfcfb9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f6d280c3aec43c49f021558f24519cb : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd3857ce66643c8923c2d9ac3a490c5 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a74c26a0b9d743dc8bef83b38aed43ed : FAILURE in 7m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cd409c3709344a6830030e2144c1600 : FAILURE in 7m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c046902f4e764923b64bf1909b01dba0 : FAILURE in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12f213947c640b1898ddd0139639faf : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/814c04bda7734e7caf5842423f35b79d : SUCCESS in 31m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2a2117b14c0406c9d7f545eb9174b46 : SUCCESS in 1h 04m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf8dd75bd8b54eee89d92019c70487a2 : SUCCESS in 1h 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d3c553205b449c82bd70cfc6350048 : SUCCESS in 1h 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed11b0fcabff44e99f16ca40f8e1cfc6 : SUCCESS in 6m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b064272d92df465faa604f0ab211e191 : FAILURE in 19m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5882ac82d124545b5863482f051382b : SUCCESS in 3m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4b4f6334f5543a685141de334c9c2ad : SUCCESS in 5m 44s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4f27cbcf402f42c8bb672306d4ab1172 : FAILURE in 22m 02s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0808914d4c0f43799a79c2cdf4cdc229 : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b95e46ffd7df49f088510a5c8d2a4ab9 : SUCCESS in 5m 33s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d2c4fa3379ff4313b52e014033d201d6 : SUCCESS in 8m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0bd8cb43e60e4c48891e7b84aed5bd59 : SUCCESS in 22m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/151bef0bb1684838a5443ac4128dc7b1 : SUCCESS in 3m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1bfdd2411f904b339980983612b170fd : SUCCESS in 1h 07m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a1186d9528245299e120a8f1946f188 : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4bbc2901cfe541f6b73a54bf468a7a42 : SUCCESS in 1h 49m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5b82cb8927af467ea15e1c4814949581 : TIMED_OUT in 3h 29m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cd292ca91d884f00a5482153d4efa208 : SUCCESS in 1h 52m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e2c1e9f1f66744ac8827cdf78ed9653a : SUCCESS in 21m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da0b4a3c4ee941cdae7451308c08c0b0 : SUCCESS in 1h 17m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8fa13f83674649f8a20663bbf10f5b3f : SUCCESS in 2h 01m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/70ad0111785146639d1e67040e92c98d : SUCCESS in 48m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9257e39a3634af0a7379077bbe6e8be : SUCCESS in 1h 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c465c57f34c4552abe7740a99f0e2e2 : SUCCESS in 34m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"491fb5b6e9c94b979eab6eeed7959492e47145e9","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 11:29:12.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/383 : SUCCESS in 0h 41m 34s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":18},{"id":"a843af377d4ad523c3171bf939a89626dd26b74d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 11:52:23.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/289 : SUCCESS in 0h 41m 47s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":18},{"id":"3a38f29e6358f404e89f6b102082d6b7ca41a46b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 12:33:32.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/655 : SUCCESS in 0h 40m 2s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":18},{"id":"d5d216763c170f96e57ca4967f5f977e410dce2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-30 12:44:18.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4579/.","accounts_in_message":[],"_revision_number":18},{"id":"d9831071fb9dcdfee6e63ad48d4a5a7339c143bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:35:09.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"dd58e6f5f20a8e6cc95803d89c8338a1344b0381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:40:04.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"b4ccc64cee22c65946ce6e60618f4f1b8b557442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:51:47.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"2624204886800f3bf817f73b93869cba143f9e59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:52:42.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"71d30928e298b754fcd579c0c5d953ff0def3de6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 14:37:37.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":19},{"id":"8b395bdecee430e73a01cf3f21d72c356e250ffa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 14:43:20.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":20},{"id":"4ec975961b2b3c54a026e9eadb0669141e34a8f6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 14:51:22.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":21},{"id":"04ea69de4ff7c3f86d680f8d1fc3e763dac64c63","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 14:55:32.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":22},{"id":"529c5502d913a18c4830492cd0da48b7994bab4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 14:59:47.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d5a30d59a724ec2811b3979c07c238b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bea427e4ce1d4633af2992af2312a78a : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2349ad2d367f4a3ab3e4653be1b1219f : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/77794249428747da82a2483eab4270e3 : SUCCESS in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6948f6408811998e7c2f13d62a6ad4ca59acca99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 17:11:54.000000000","message":"Patch Set 22: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea5c649e6d6c4e66bfa91ecc0fba4dd1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8897133da30e4916b0f5574d8f7123e8 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6ece8c6b24474796d3e2394c8a1f80 : FAILURE in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65416ab2f7c24881bae32f12a3256227 : SUCCESS in 8m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67654c6b849a44898fea810386ae83cb : SUCCESS in 7m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/804ad58bc91944f6b9f97bbb79c0fdf2 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba304caa75d4f3fb2e10329e3370757 : FAILURE in 6m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/4b9e1ed6236f446f92d8ca194e5d65bf : SUCCESS in 54m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52ebbb338f60477a800c64bcab0d3137 : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fe0f63ea31949b58dbc498641ac990a : SUCCESS in 1h 33m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/379793340e254447a0c66ef478fbcf8c : SUCCESS in 1h 06m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe3088fdf9545698b27762ea4aabc06 : SUCCESS in 9m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7409f171fdc941eb8e4ee8222ccfc87c : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be4910ecf25a463bacfd51c89a42fe6c : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69425c2262d04914871a22f901db48d3 : SUCCESS in 3m 56s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4d71174f83354e1aaf3dfa10ba7c9af3 : FAILURE in 22m 31s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/00356faa9efb4f78a239f16de4efc69c : SUCCESS in 6m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eaf0c4422a44446085c873b144a9c98e : SUCCESS in 6m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d22552a81e65477f9bffb071440f5567 : SUCCESS in 8m 19s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/46d2421f9a6c453fa9165c70d92ed350 : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed97dfbf350143e1bf83b450b337e5e7 : SUCCESS in 3m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/000f6620568b4b50bba93c0c0b94761f : SUCCESS in 1h 26m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/285828032d86419598baa0965ef230b7 : FAILURE in 2h 21m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e8964349ea1c42b68bc3887c503f2cbf : SUCCESS in 1h 52m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4d7e766cfcd34ce7b19ea76e0a1c1fe3 : FAILURE in 3h 05m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/eb8446c042ac4d96b790546edacea90a : SUCCESS in 1h 59m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/433cfd18e5c84cb2bc877ff191b1af69 : SUCCESS in 41m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8f92d5df78b439983bd0e7dcf804a76 : SUCCESS in 1h 27m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ee4e3becbbd04f2ab17a1812a183739c : SUCCESS in 1h 39m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea9c382a69964f56bc340014e8a610e4 : SUCCESS in 52m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3210a80907fb4c4eafd956150e9905d3 : FAILURE in 1h 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/678e4b9a4942440180781f519b7946f4 : SUCCESS in 1h 03m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"eff93b90204787c24bd59eb790d842809d2f60e8","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 20:37:52.000000000","message":"Patch Set 19:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/303 : FAILURE in 0h 31m 23s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":19},{"id":"371accd708826e8283da6affb694af0dde5da251","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 21:15:59.000000000","message":"Patch Set 20:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/304 : FAILURE in 0h 31m 22s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":20},{"id":"477776b9a9d5c7ef1b29e74f07aa5162cb3f2336","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 21:33:37.000000000","message":"Patch Set 19:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/397 : SUCCESS in 0h 42m 12s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":19},{"id":"56c749403e52adbae7d787f47042db45e3e8a75f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 22:21:20.000000000","message":"Patch Set 20:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/398 : SUCCESS in 0h 41m 18s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":20},{"id":"5d43e4cf93d883f2e5423fb72bd18b2979fb9d43","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 22:47:25.000000000","message":"Patch Set 19:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/669 : SUCCESS in 0h 41m 44s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":19},{"id":"649b710ded65c85dd1497e92a5c95a3f2dfbceb6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 23:06:16.000000000","message":"Patch Set 21:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/306 : FAILURE in 0h 29m 54s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":21},{"id":"53464d2cbc6321da65013867ad010cba784f59df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-30 23:31:04.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1c4e6bc4268d40f0aeaa617e83aecbbd : SUCCESS in 50m 32s\n- cinder-storpool-tempest-iscsi-multipath build/3b9687d39458450e91995c3fad14c131 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":22},{"id":"98063235934083e66ff569b48f067bdfec5dc8b6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 23:35:10.000000000","message":"Patch Set 20:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/670 : SUCCESS in 0h 41m 10s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":20},{"id":"78620b9c7a4d92556686289a5dd4de1b7e7db63e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 23:43:40.000000000","message":"Patch Set 22:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/308 : FAILURE in 0h 30m 53s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":22},{"id":"97ca9af2af02b07fec0538e9c2fa191be99ac676","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 23:58:00.000000000","message":"Patch Set 21:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/400 : SUCCESS in 0h 41m 40s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":21},{"id":"87e688c5dd754e333cf953c869e2fdef349239eb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 01:10:40.000000000","message":"Patch Set 21:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/672 : SUCCESS in 0h 40m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":21},{"id":"0512916c5dbd29b9723a71545b19bd462240a688","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 01:12:40.000000000","message":"Patch Set 22:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/402 : FAILURE in 0h 31m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":22},{"id":"d7c93d88c20950c3bea172fe1e0d5acbb171733e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 02:35:39.000000000","message":"Patch Set 22:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/674 : SUCCESS in 0h 41m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":22},{"id":"3cf417846f6a2ac143edda4eeadc9d9325f6b57f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 04:01:52.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4593/.","accounts_in_message":[],"_revision_number":19},{"id":"4bfb8ce468320a177c1b532cfa523a9ca8f42d76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 05:04:17.000000000","message":"Patch Set 20:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4594/.","accounts_in_message":[],"_revision_number":20},{"id":"0106b724f9dbade0b92b8db401fda0cd07179ae4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 07:14:42.000000000","message":"Patch Set 21:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4596/.","accounts_in_message":[],"_revision_number":21},{"id":"1f10761f23f4d207af06d404535c741760d29eb7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 09:42:55.000000000","message":"Patch Set 22:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4598/.","accounts_in_message":[],"_revision_number":22},{"id":"af5fe6de28868e79708f0a974e8ed4181bb14d78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 12:03:50.000000000","message":"Patch Set 16:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4291/.","accounts_in_message":[],"_revision_number":16},{"id":"081a6778eb4edbecde53de33a23389266c331fad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 10:18:55.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4400/.","accounts_in_message":[],"_revision_number":17},{"id":"70cc02244ae35c6fdb0ee2c906b2b4d6bedf8149","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 10:46:50.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4403/.","accounts_in_message":[],"_revision_number":18},{"id":"ab9854d88f72d5a16eb04e2f500215f43a2ee195","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 13:47:43.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4417/.","accounts_in_message":[],"_revision_number":19},{"id":"215c17b64457eb2adcb2081ae4d3f3489893a9a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 13:57:56.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4418/.","accounts_in_message":[],"_revision_number":20},{"id":"cf2abd2de99a706619dddda0e16efa5d26f44916","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 14:21:46.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4420/.","accounts_in_message":[],"_revision_number":21},{"id":"e905d2485816cef2323582995415f197b15af7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-03 08:55:57.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"d5a8c933d28ef78d707fbae82fc4004e6d588d72","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 09:06:48.000000000","message":"Patch Set 23:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4823/.","accounts_in_message":[],"_revision_number":23},{"id":"d9af51839ac2cb76b91b5a4a3b8c19bd7c562efb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 09:08:31.000000000","message":"Patch Set 23:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4999/.","accounts_in_message":[],"_revision_number":23},{"id":"c9c8e33240ae28a96e9dd93b99723a2dd27509eb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 09:32:10.000000000","message":"Patch Set 23:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a1699900265a492ba3bf539081add8ee/","accounts_in_message":[],"_revision_number":23},{"id":"7db4905ac92ce19eb214a714b7b2486dde76d10a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-03 09:44:38.000000000","message":"Patch Set 23:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1058 : SUCCESS in 0h 47m 45s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":23},{"id":"6e62ec5c9616f878c2d88b891e585edc6aed094b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 09:47:12.000000000","message":"Patch Set 23:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a1699900265a492ba3bf539081add8ee\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf3b2f1390e94bd79eff4268a6104e18 : SUCCESS in 49m 19s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9c78a0c93464364b5848bc67fd1d78a : SUCCESS in 41m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a86942c4e7f043ce9ab638b4b76e826b : SUCCESS in 48m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55ffb592d89b425cb0e0b7e11e7b8522 : SUCCESS in 35m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ce421902c4040caacfeeeadd7b0347c : FAILURE in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"fff9e5522de095fd4838760462d4a77cc3f33cfc","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-03 09:47:27.000000000","message":"Patch Set 23:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/790 : SUCCESS in 0h 50m 33s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":23},{"id":"c4712134eca35af69e930916747bf9d54408c481","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-03 09:47:38.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e35c960616db4fc3bc7510572336e1de : SUCCESS in 46m 31s\n- cinder-storpool-tempest-iscsi-multipath build/61f4a9549b9d49a3978daad1c9daf20d : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":23},{"id":"e42e5718dd362c9c35a18430303f630111c91522","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-03 09:47:39.000000000","message":"Patch Set 23:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/697 : SUCCESS in 0h 50m 44s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":23},{"id":"aa8ab9a264df3039ff66b542565b82bc68ef649c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 09:58:44.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":23},{"id":"f7ab563e00932e17b785af4d36fff08ae1dd8585","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 10:05:11.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7536cc0ebf614af4b24dbd08d1966576\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28ec93113f9349fcbba08bace12ec92d : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bcda5378297046afbb693e0813f942f8 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/82097cda1b144c5b97cfd2f4d02ba419 : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"daa6d17168ce65a3c2f221d7b25d68411c288e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 12:05:00.000000000","message":"Patch Set 23: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d951288513a14715bd864137bf58bee2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3712606d39cc477d94bb69510cf9ea0d : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f226a12f9043f095845e0041057693 : FAILURE in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0422ee21b3044ebc9498576048060138 : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f93154e7659f474a9e23d9698ef3ef71 : SUCCESS in 8m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/674ccbf1eab0416e8fc9e8b588edf1ce : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01849a71b3b4ecd8433fae1961a78d9 : FAILURE in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d037c91f15354ab0a6d26177f6cd85f0 : SUCCESS in 1h 09m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24536bf80ecc410ca7147cc072b5dbf5 : SUCCESS in 1h 03m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c97a8af3991445dfad814602d3c2106c : SUCCESS in 1h 29m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a35bcdeff8254ff6b7aab19dcaea5de6 : SUCCESS in 27m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0009bf76a84046daaed91da14218e82a : SUCCESS in 7m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a98202295bd442579ef22b7270308605 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/34d82dae693a4b92bc3e376a4d4aa91f : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f9c58fd54264f51b9aabb463e7c086d : SUCCESS in 5m 29s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/67c5a97b6ea24ee08ea3a0cfc2b12721 : FAILURE in 13m 11s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/337679e7f9f841df8faf59998d5d0b12 : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a6ab72729f6046289b5371bcfa8aacb7 : SUCCESS in 5m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9de1ba74237b44e3b7ce1462c6459316 : SUCCESS in 9m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/86cb394a1dd14d258cc6b79fe32e6942 : SUCCESS in 37m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5f870c4ae7949e892c782b0ae50fc10 : SUCCESS in 3m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/501b78651e444f5399e065213048b149 : SUCCESS in 1h 30m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ce9b5e4f2eb43da9dcb7a5e237d806f : FAILURE in 1h 54m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cca1732f5788445aa273fe5d206d1116 : SUCCESS in 1h 52m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7f68ba4a205a456480496ce5d1dc51d6 : TIMED_OUT in 3h 07m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e75946bc2d51476e8a0afd82eddb6134 : SUCCESS in 1h 57m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8054ed4ce3c9498292e0a91ec871052a : SUCCESS in 17m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f8d9cf4e95b45dcbb7d888ae168339c : SUCCESS in 1h 12m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/feee31ca6f6345b9aec9e3e74f92b5b6 : SUCCESS in 1h 12m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b0cad53afae46baad6a7ee5bbd3e431 : SUCCESS in 25m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f3990639e484f08b456186589c3500e : SUCCESS in 39m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e143bb3d20044619d9b5debc3db6337 : SUCCESS in 29m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":23},{"id":"bde32b4c4ab9817519fd5adf94c10071682c74d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-04 07:08:42.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"2dbff9dc9edfbc73fa31baa2b7bc38ecd44816a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 07:20:23.000000000","message":"Patch Set 24:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5009/.","accounts_in_message":[],"_revision_number":24},{"id":"57ec1ac96e0561e16cf1747edfbeb6eff9e72848","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 07:20:30.000000000","message":"Patch Set 24:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4833/.","accounts_in_message":[],"_revision_number":24},{"id":"8e73dbd098c0e69c5f403782152f7c0d62072343","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 07:49:37.000000000","message":"Patch Set 24:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ae4d856fe07f480181a8b74f9f98a5de/","accounts_in_message":[],"_revision_number":24},{"id":"9df8fb22dd570a5fb242f7a37f30037bd0a9129f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 08:07:24.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I69f0469b1a6c543752502f2f915001caf4570686 .","accounts_in_message":[],"_revision_number":24},{"id":"19ed5347d1bf4d9956ddb38d80edb87bf5af8728","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 08:11:33.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e37b739a774a14ae9deff0bc2902bf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ffe084e9e064993897b42598591c7bb : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e222cd53a7b5452c86d22cc796ede9ad : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4141f8242ee465ab5fe7bc5a80b80b1 : TIMED_OUT in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ba0437f522ee0a5f93048cc7249237313177d5c1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 08:19:00.000000000","message":"Patch Set 24:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ae4d856fe07f480181a8b74f9f98a5de\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/066692ad82564f9a93fa61e4208e063f : SUCCESS in 39m 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/10909768989b4906a570baba35efc85e : SUCCESS in 40m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26e1dd39e26d4972b36744d5e66fa354 : SUCCESS in 43m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d2d0f39df36740d38a471c81d0a26261 : SUCCESS in 42m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03dc058f77f44d998d17a37fd2faa064 : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"12197f8d6585db4e789fd862ff766f6836545a9f","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-04 08:21:07.000000000","message":"Patch Set 24:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/705 : FAILURE in 0h 37m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/928846","accounts_in_message":[],"_revision_number":24},{"id":"d15a771bb441cd6f05db34f580b5f4e733cc9006","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-04 08:22:13.000000000","message":"Patch Set 24:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1066 : FAILURE in 0h 38m 0s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/928846","accounts_in_message":[],"_revision_number":24},{"id":"b072c574bc88afd7b24f0f72c77c3686628410b5","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-04 08:23:41.000000000","message":"Patch Set 24:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/798 : FAILURE in 0h 38m 50s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/928846","accounts_in_message":[],"_revision_number":24},{"id":"23db916239ffcde58086e01d4c7142c78bfe98bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-04 08:31:02.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/12eadfef846747ee8622b1d7ed52df3c : SUCCESS in 48m 50s\n- cinder-storpool-tempest-iscsi-multipath build/7bb4e39171ce4c4cb708dcd1f1752e29 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":24},{"id":"849eaee67dda54c9d4ffcc4df754da99a6e18ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:14:01.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61011ab9412e431a89c5ae021ea837d2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/46544fca51a74e83830a98bbaef2ba62 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0624bd87654f4cb597cdabc36b97b131 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b564901e0984cc4aeb6450770e476a9 : SUCCESS in 10m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d392b3ff45b40b5bcd83b0050c6a828 : SUCCESS in 8m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a681264d1bfc46b484460dac9b67a111 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1a7d4918f54c48ac90f2ea38904bd1 : SUCCESS in 14m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/68267a323c654de6848695211a9f6f21 : SUCCESS in 1h 06m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40c2a656ec8f4ece951ed7eb545ab5dd : SUCCESS in 28m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8e94d3ecc5c4e5c885f85078501053e : SUCCESS in 1h 21m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1096d987d02e4371a0e07f1c31470c4c : SUCCESS in 33m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378bc014367c42bf96b13b63f1549a6e : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2c7381643a824ddfaaf5e98001542607 : SUCCESS in 19m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8c2774394ef244f9956c859644a795fc : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17f933f2614140d7a83ff3a5f2b988c0 : SUCCESS in 6m 27s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/0024489c96ba4c80a50ab4f9cf15975b : FAILURE in 8m 46s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8adc9b729cbc4332bab68eb294cdb92e : SUCCESS in 6m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b495fed0002f41f899da5524ffc0e1d7 : SUCCESS in 7m 31s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7833a9eab8a84eeda905676aef1b8962 : SUCCESS in 7m 10s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ddee630262a4968b3b1d4c681d9f5db : FAILURE in 44m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bfaad4eacaee4da8982a00eb16f27895 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9faa7e1da5b424183892642f9e1b64c : SUCCESS in 1h 18m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab755cc3320141bbac32262a781a023b : FAILURE in 1h 57m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/663bfa8065b84644aa36f1d5bf531658 : SUCCESS in 1h 21m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/82bb24067cf94291939007806ec1af44 : FAILURE in 2h 59m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a2fadb916d6e491cb4056baf0ecfd9e3 : FAILURE in 1h 44m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6f5e9482bc7145db82fbfbbddec888f9 : SUCCESS in 19m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c744362d7b424d06a8f3963fc17c7f67 : SUCCESS in 1h 14m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/252866443e9f4fe1bd126f9ef1b4243e : SUCCESS in 1h 48m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3bb500e43bad4a738ef5558c354e4df3 : SUCCESS in 45m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/858fa357122e408f90ad6debbd282653 : FAILURE in 1h 50m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63c1a7065a4a434994ed5997e02de6fc : SUCCESS in 31m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"ed67f532253031b627156022a7ec0599e7d41cfc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-04 11:44:56.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"f48fa4d4bdc6e3899ef6b747ea8396ca13514e89","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-04 11:59:59.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- cinder-storpool-tempest build/2863ab6123864f1da2c5e35dcc2ee83e : RETRY_LIMIT Ansible setup timeout in 1m 23s\n- cinder-storpool-tempest-iscsi-multipath build/e931b4f737d045dbb9aac67a5a118712 : RETRY_LIMIT in 8m 40s","accounts_in_message":[],"_revision_number":24},{"id":"da5275a0e2b2f592b792e982c32167946739bb3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 12:12:58.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9843d99fc3774932ac078d695225d892\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98dfbd59239e40f2ade1787c3a13f8eb : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3c7080b5490451f935b2b7480276b56 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/885a6b6205394740b99386a3ccf67cfa : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0a963fd4fd603de7c85b43912db7537d61d5ccc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:52:43.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65a91b0d42994e1ebc8b09f2dd69daad\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f8bf84192edf4d99b50eb40bc127148c : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f91adb9a504a58bf52d114465f3426 : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a0947eb0e7144ddb9970527fed8222a : SUCCESS in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8041a2ba472a43358001b50b768782fb : SUCCESS in 8m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/006f56f8e683460a925d3f10734b25e5 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c4d1d94cb74bd59b7501331855e1d2 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ed85cb27c2d147eb9b15aaead955c7ca : SUCCESS in 31m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e86d67486f19409386b8bea173808d6b : SUCCESS in 38m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1db8480f99c4633bd5786e9ced7b0b1 : SUCCESS in 1h 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90dcb5213e0b48e189906e35c5c12f4c : SUCCESS in 29m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd1cb56719d4403885ba032a05e4e97e : SUCCESS in 6m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8d551dfef8644f77a95086c946b5b626 : SUCCESS in 27m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cc62398c7074a608234b16e574f1426 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/170e3d091a954cbaa67e150206bbb18e : SUCCESS in 6m 01s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a5fc0046094c4cfe894a31c40905b999 : FAILURE in 13m 18s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f46bdd19d3694ff78ff45f0d40f28b64 : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3a7e36abde3746faaabe1416a08b437b : SUCCESS in 5m 26s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d6e01283d3ac4e71b051476879b344e5 : SUCCESS in 8m 48s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ee3a3d451ef4a8aaeacc52df82ad441 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62cce5a302d441a98824cafe5f1cbb2f : SUCCESS in 8m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3298bc6202e439ebb67b954620d06d4 : SUCCESS in 1h 08m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d50e623a10ef49ec9454f9bdc0350ef6 : SUCCESS in 1h 32m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1c956499dab496c9bc5485be01cf081 : SUCCESS in 1h 14m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6459ee4aa0c244159987fae60f493d11 : FAILURE in 3h 05m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b0c5fb311aac4175920b5d708e1a9ddc : SUCCESS in 1h 49m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/933b96e119584100ad30a7cf6df4925a : SUCCESS in 17m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/577c1106067c40038a4b307d30d91026 : SUCCESS in 1h 06m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cc1e214436b34146b040fb27c65ef265 : SUCCESS in 1h 34m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9e6ce77c56f4d4bb777bb4182cc8d9e : SUCCESS in 39m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6404dd69fcbe4adb84582873597e7e2c : SUCCESS in 1h 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e6be316390f43bbb2a7ac575d1d8d21 : SUCCESS in 54m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"2e80591a8189428d3e477e3065108e4c61f987d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 15:03:29.000000000","message":"Patch Set 24:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_08_2026_14_56_39\nLeave a comment with","accounts_in_message":[],"_revision_number":24},{"id":"f1f8aa87332e3e2b3e2f2902c8bc54aba5997eff","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-18 12:54:34.000000000","message":"Patch Set 24: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"2959ed5daa4470f63cc61b74c5dce98888a64e6c","revisions":{"44e387d59ae3760e233a441c6ddaf686bdf315d7":{"kind":"REWORK","_number":1,"created":"2024-09-10 17:51:05.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/1"}}},"commit":{"parents":[{"commit":"8dce75e1d0752134b90b30212678665170f1804d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dce75e1d0752134b90b30212678665170f1804d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:50:46.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44e387d59ae3760e233a441c6ddaf686bdf315d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44e387d59ae3760e233a441c6ddaf686bdf315d7"}]},"branch":"refs/heads/master"},"25093e5692618e9a03930e2ba751abf2b154cc9c":{"kind":"REWORK","_number":2,"created":"2024-09-10 20:58:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/2"}}},"commit":{"parents":[{"commit":"8dce75e1d0752134b90b30212678665170f1804d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dce75e1d0752134b90b30212678665170f1804d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 20:58:40.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25093e5692618e9a03930e2ba751abf2b154cc9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25093e5692618e9a03930e2ba751abf2b154cc9c"}]},"branch":"refs/heads/master"},"1d9f93f2327295b0507b94a128114057affdd3fe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-14 11:00:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/3"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 11:00:28.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9f93f2327295b0507b94a128114057affdd3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9f93f2327295b0507b94a128114057affdd3fe"}]},"branch":"refs/heads/master","description":"Rebase"},"9c000acf24df4ff3c0689bda7c017b958155edf6":{"kind":"REWORK","_number":4,"created":"2025-04-14 11:01:08.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/4"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 11:01:07.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c000acf24df4ff3c0689bda7c017b958155edf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c000acf24df4ff3c0689bda7c017b958155edf6"}]},"branch":"refs/heads/master"},"b5b05a8665ce76f1956dca4e126dbfe53de2bb3b":{"kind":"REWORK","_number":5,"created":"2025-04-14 12:55:18.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/5"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 12:55:08.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b05a8665ce76f1956dca4e126dbfe53de2bb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b05a8665ce76f1956dca4e126dbfe53de2bb3b"}]},"branch":"refs/heads/master"},"e144e4a153162e0c937484b217cb8699a366ed88":{"kind":"REWORK","_number":6,"created":"2025-04-14 12:59:23.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/6"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 12:59:16.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e144e4a153162e0c937484b217cb8699a366ed88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e144e4a153162e0c937484b217cb8699a366ed88"}]},"branch":"refs/heads/master"},"3d790c8704e95a6a76566eba457e415977b4550e":{"kind":"REWORK","_number":7,"created":"2025-04-14 13:01:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/7"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 13:01:52.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d790c8704e95a6a76566eba457e415977b4550e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d790c8704e95a6a76566eba457e415977b4550e"}]},"branch":"refs/heads/master"},"7f809e8c5641415649c41e7a699670cf5bd98fa2":{"kind":"REWORK","_number":8,"created":"2025-04-15 06:04:48.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/8"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-15 06:04:26.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f809e8c5641415649c41e7a699670cf5bd98fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f809e8c5641415649c41e7a699670cf5bd98fa2"}]},"branch":"refs/heads/master"},"df62f8e8ff3760b981e5d443b170e28cf6e7eb3a":{"kind":"REWORK","_number":9,"created":"2025-08-29 07:00:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/9"}}},"commit":{"parents":[{"commit":"644b6362a6b0debf6395d4bf15f963faf1a42ced","subject":"Merge \"mypy: enable strict equality checking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/644b6362a6b0debf6395d4bf15f963faf1a42ced"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-08-29 07:00:49.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df62f8e8ff3760b981e5d443b170e28cf6e7eb3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df62f8e8ff3760b981e5d443b170e28cf6e7eb3a"}]},"branch":"refs/heads/master"},"184b5f5f2400564c53cd541e95cd7ae9ac9cbe3a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-23 07:59:29.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/10"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-10-23 07:59:06.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/184b5f5f2400564c53cd541e95cd7ae9ac9cbe3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/184b5f5f2400564c53cd541e95cd7ae9ac9cbe3a"}]},"branch":"refs/heads/master"},"26fbb8c234981a8b5351711a6a8f15bc44e9b99e":{"kind":"REWORK","_number":11,"created":"2025-11-03 15:51:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/11"}}},"commit":{"parents":[{"commit":"c9f569a4e434bef7f3a7a442ddaeba6621100732","subject":"Merge \"[NetApp] Default node model to empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9f569a4e434bef7f3a7a442ddaeba6621100732"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-03 15:50:52.000000000","tz":60},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26fbb8c234981a8b5351711a6a8f15bc44e9b99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26fbb8c234981a8b5351711a6a8f15bc44e9b99e"}]},"branch":"refs/heads/master"},"ef1faa4d19e1fc6c4eea1561dd49c4ae709a9554":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-12-16 15:26:18.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/12"}}},"commit":{"parents":[{"commit":"322d5d492d7af20354ab0a03a418dfced30adda4","subject":"Merge \"api: Add extension_is_loaded decorator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/322d5d492d7af20354ab0a03a418dfced30adda4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-12-16 15:26:18.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef1faa4d19e1fc6c4eea1561dd49c4ae709a9554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef1faa4d19e1fc6c4eea1561dd49c4ae709a9554"}]},"branch":"refs/heads/master","description":"Rebase"},"3c6a920876d539531131e8c1c403aa32d207fe4f":{"kind":"REWORK","_number":13,"created":"2026-02-18 19:53:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/13"}}},"commit":{"parents":[{"commit":"322d5d492d7af20354ab0a03a418dfced30adda4","subject":"Merge \"api: Add extension_is_loaded decorator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/322d5d492d7af20354ab0a03a418dfced30adda4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-18 19:52:47.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c6a920876d539531131e8c1c403aa32d207fe4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c6a920876d539531131e8c1c403aa32d207fe4f"}]},"branch":"refs/heads/master"},"62ca02fff7a5fa944bc1efb50ddb252e76e04f48":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-02-18 19:53:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/14"}}},"commit":{"parents":[{"commit":"347b871e44d8f0556f2be3e0967294b6394febed","subject":"Merge \"Doc: remove python 2 references, no longer relevant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/347b871e44d8f0556f2be3e0967294b6394febed"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-18 19:53:09.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62ca02fff7a5fa944bc1efb50ddb252e76e04f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62ca02fff7a5fa944bc1efb50ddb252e76e04f48"}]},"branch":"refs/heads/master","description":"Rebase"},"23dc8922b5e379e5e87c9ae68b6a9b02695767fd":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-02-20 14:11:37.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/15"}}},"commit":{"parents":[{"commit":"2f43a2c058bbad3a385b258db46cd2b357f72eed","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f43a2c058bbad3a385b258db46cd2b357f72eed"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-20 14:11:37.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23dc8922b5e379e5e87c9ae68b6a9b02695767fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23dc8922b5e379e5e87c9ae68b6a9b02695767fd"}]},"branch":"refs/heads/master","description":"Rebase"},"64b9b240f6673015dbb173e62b415a5d54c0ae07":{"kind":"REWORK","_number":16,"created":"2026-06-12 22:04:30.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/16"}}},"commit":{"parents":[{"commit":"9bfe91494e77266466064e6b633d1205a055b27a","subject":"Merge \"Dell PowerFlex : Fix multiple broken functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9bfe91494e77266466064e6b633d1205a055b27a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-12 22:03:43.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b9b240f6673015dbb173e62b415a5d54c0ae07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b9b240f6673015dbb173e62b415a5d54c0ae07"}]},"branch":"refs/heads/master"},"3a327db446bd8bf65f6ef9faa508d5de37107fd2":{"kind":"REWORK","_number":17,"created":"2026-06-30 07:20:03.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/17"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 07:19:48.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a327db446bd8bf65f6ef9faa508d5de37107fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a327db446bd8bf65f6ef9faa508d5de37107fd2"}]},"branch":"refs/heads/master"},"06a510feb8c4ccd15fd413babf2d24d8b1c8011f":{"kind":"REWORK","_number":18,"created":"2026-06-30 07:39:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/18"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 07:39:00.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06a510feb8c4ccd15fd413babf2d24d8b1c8011f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06a510feb8c4ccd15fd413babf2d24d8b1c8011f"}]},"branch":"refs/heads/master"},"9655799219517ab970444d1f5f7d17f067b24938":{"kind":"REWORK","_number":19,"created":"2026-06-30 13:35:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/19"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:34:59.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9655799219517ab970444d1f5f7d17f067b24938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9655799219517ab970444d1f5f7d17f067b24938"}]},"branch":"refs/heads/master"},"885f9c235fb8e9f7182947ad04d930e46a9beb5c":{"kind":"REWORK","_number":20,"created":"2026-06-30 13:40:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/20"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:39:53.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/885f9c235fb8e9f7182947ad04d930e46a9beb5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/885f9c235fb8e9f7182947ad04d930e46a9beb5c"}]},"branch":"refs/heads/master"},"dcd17c191ed2123fa9398a779e8ceccfb5db5d66":{"kind":"REWORK","_number":21,"created":"2026-06-30 13:51:47.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/21"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:49:49.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcd17c191ed2123fa9398a779e8ceccfb5db5d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcd17c191ed2123fa9398a779e8ceccfb5db5d66"}]},"branch":"refs/heads/master"},"751e2624a5ec86197435f04ba88f8a3cf440f7f1":{"kind":"REWORK","_number":22,"created":"2026-06-30 13:52:42.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/22"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:52:25.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/751e2624a5ec86197435f04ba88f8a3cf440f7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/751e2624a5ec86197435f04ba88f8a3cf440f7f1"}]},"branch":"refs/heads/master"},"0e2e2d4a40483776e9a52bde70e564c568630f65":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-08-03 08:55:57.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/23"}}},"commit":{"parents":[{"commit":"316c908ca8f2a16c2a9f1a5bf66859997797713a","subject":"Merge \"Tests: Mock sleep in fungible unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/316c908ca8f2a16c2a9f1a5bf66859997797713a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-08-03 08:55:57.000000000","tz":0},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e2e2d4a40483776e9a52bde70e564c568630f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e2e2d4a40483776e9a52bde70e564c568630f65"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","ours":"751e2624a5ec86197435f04ba88f8a3cf440f7f1","theirs":"316c908ca8f2a16c2a9f1a5bf66859997797713a","merge_strategy":"recursive","contains_conflicts":false}},"2959ed5daa4470f63cc61b74c5dce98888a64e6c":{"kind":"REWORK","_number":24,"created":"2026-08-04 07:08:42.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/46/928846/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/928846/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/928846/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/928846/24"}}},"commit":{"parents":[{"commit":"1736706bddccd2732374a2cee40ed7f922501112","subject":"Merge \"Fix Google backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1736706bddccd2732374a2cee40ed7f922501112"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-09-10 17:49:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-08-04 07:08:34.000000000","tz":120},"subject":"Add backups summary API","message":"Add backups summary API\n\nThis adds a backups summary API in, the same way\nthat the volumes summary API is implemented, in\na new API microversion.\n\nThis is a good way for both end-user and\nadministrators to get an overview of the\nbackups that is stored and it really helps\nwhen having thousands of backups to get an\noverview on the state of the backups\nacross one or all projects.\n\nCloses-Bug: #2107329\nChange-Id: I69f0469b1a6c543752502f2f915001caf4570686\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2959ed5daa4470f63cc61b74c5dce98888a64e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2959ed5daa4470f63cc61b74c5dce98888a64e6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
