)]}'
{"id":"openstack%2Fcinder~929701","triplet_id":"openstack%2Fcinder~master~I116dc49ebf2323480278c308ec2635fc866b5d7e","project":"openstack/cinder","branch":"master","topic":"update/cinder-docs","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2025-02-24 20:54:37.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-02-24 18:10:17.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"36569":{"account":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"last_update":"2025-02-24 20:54:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I116dc49ebf2323480278c308ec2635fc866b5d7e","subject":"Updating the Lightbits Cinder driver documentation regarding renaming.","status":"MERGED","created":"2024-09-17 19:40:18.000000000","updated":"2025-02-24 20:55:52.000000000","submitted":"2025-02-24 20:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929701-update/cinder-docs","meta_rev_id":"a70ecaaed9611772b441634493f990d3ea85f7a9","_number":929701,"virtual_id_number":929701,"owner":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-24 20:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-02-24 18:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-02-24 18:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 19:40:25.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-17 19:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 20:35:07.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-17 20:36:37.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-17 21:44:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-09-17 21:49:47.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-09-17 22:09:20.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-17 22:23:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-09-17 22:28:26.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-09-18 00:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-09-18 01:17:48.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-09-18 02:56:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-09-18 11:25:25.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-18 15:19:23.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-09-19 06:35:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-26 10:02:20.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-09-27 07:01:51.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-10-06 15:07:13.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-10-11 18:14:57.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-12 18:18:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-01-18 05:50:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-01-18 07:05:23.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c97c40b494b92f1a39159136abd022099456605b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-17 19:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45a60d720977598aaba74b3991e70ede25c2b03c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 19:40:25.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/567/","accounts_in_message":[],"_revision_number":1},{"id":"2064a50d3ecd1700e70a7af5acd0ec1ef1ea0668","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 19:40:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"a3ee9434bc1c28c03abf235394de8fc3ef57e3e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-17 19:42:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2f3bda4d23bd418b35f64a4c0382744b7dd033c5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 19:42:55.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/568/","accounts_in_message":[],"_revision_number":2},{"id":"b72b852016f7708f1b61e65845d12003a131b889","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 19:42:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"c9572d678acd6f65449c0b15313014cb0607e3f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 19:56:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8f250d34ad44f18b330627761714cda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5710e95778fe44c395bc2c389cc42160 : FAILURE in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d7fef7e46b42d78e361ea547683955 : FAILURE in 6m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c9df9e5fe144b108ab7e14a13d10a04 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b60cc819649605fd82ee7111fe1d9436fc9910b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-17 20:35:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/929701 : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3109e1452d442c0346cfd636d79fc1f0068cbbe4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-17 20:36:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/929701 : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":2},{"id":"661e077995c14d37cb3f3856ab25b237c251c5e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-17 20:58:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"756e0ad4ed448a32b303dc9d3f6b2d608846040c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 20:58:40.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/569/","accounts_in_message":[],"_revision_number":3},{"id":"92cef1d5254de927ba822d603c46630b78c40170","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 20:58:43.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"05dbe55c769a404a8fe497ea92c0f9f9eaf0bc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 21:39:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a46b03aa4bb44b8863cf380f6b53c20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44659736a664b11b84ec3898e4cc954 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96aaaefda3c84813aaad1b26f5358070 : SUCCESS in 20m 39s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f3ac31f2f1444e28820ace2371fd109 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":3},{"id":"80fbcea103deda4829e420119241246c749b05f9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-17 21:44:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/929701 : FAILURE in 44m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0435a74d66ec9717dc812d3d134811b1cbb97bcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-17 21:44:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2024_21_39_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"5dd4a051a5d118bb4bcbea606cadee8b37e7aaf9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-17 21:49:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/547/.","accounts_in_message":[],"_revision_number":3},{"id":"404e0a5375c5aac3945eca95f3f40e85c707b134","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-17 22:09:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/929701 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b6c57a580c126eff9bd17ad53bf4ea199c997832","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-17 22:23:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"208779b2acd573bd32279c799d2d4937a5c6aa1e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-17 22:28:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/929701/3/upstream-check/cinder-tempest-plugin-ontap-nfs/3532166/ : SUCCESS in 47m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"db94ee375fb98afc40bfbf9e8099cb61b0281cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-17 22:54:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"92e0fa20a78fadd11580a276161041a8f2d36a3c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 22:55:05.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/570/","accounts_in_message":[],"_revision_number":4},{"id":"7160663654cf653c8e23106fd65af16c5297c610","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-17 22:55:08.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"ee66301adc853b3c0d268816f7c105635b8d921d","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-17 22:56:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4e73dff402e7db41b36388a898c11dfa204c913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 23:07:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfeeb66c7e3e45cdbcbbd48daa80e4ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b81914447894e97abda7177bca01d53 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac33967f5e540018fc55a4467ff6a59 : SUCCESS in 11m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/26689641c3ba49eb9433e5d62001b44b : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e119085ce6fb184173064d32ccb94e82948db33a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-17 23:12:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2024_23_07_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"5bec3612287aab013c9133f0d5f23a8ebe0b0928","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-17 23:40:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/929701 : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":4},{"id":"70496bd28002422120cea1a02ba87995395e3201","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-17 23:45:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/548/.","accounts_in_message":[],"_revision_number":4},{"id":"61a71f05581b0d8b2081c7b7e564ecadc617db16","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-17 23:46:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/929701 : FAILURE in 40m 23s","accounts_in_message":[],"_revision_number":4},{"id":"d8da3783a3eebb7014cca3410fcf41057049cbb9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-18 00:00:10.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/01/929701/2/fc : SUCCESS in 0h 42m 45s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/01/929701/2/iscsi : SUCCESS in 0h 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9e1024024546c6ae875707c4722f61308c5a8ff0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-18 00:06:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/929701 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a0274d09bb31f0b25932c3afe757111f6f391121","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-18 00:06:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/929701/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/572d539/ : SUCCESS in 47m 39s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/01/929701/4/upstream-check/cinder-tempest-plugin-ontap-nfs/4aeafc5/ : SUCCESS in 46m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"43d6fcb613209301dba240b7faa6160773102f28","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-18 00:31:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1137/.","accounts_in_message":[],"_revision_number":4},{"id":"c93d44589b6df4df7096f5a40bea483f630266d2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-18 01:17:48.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerStore_iSCSI/4825e4e/DellEMC_PowerStore_iSCSI/1743 : SUCCESS in 1h 21m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerStore_FC/af65c37/DellEMC_PowerStore_FC/1941 : SUCCESS in 1h 04m 40s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerStore_NFS/a80b72f/DellEMC_PowerStore_NFS/1436 : SUCCESS in 1h 13m 19s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerStore_NVMeTCP/e95f64d/DellEMC_PowerStore_NVMeTCP/1296 : SUCCESS in 2h 10m 29s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerStore_NVMeFC/5729661/DellEMC_PowerStore_NVMeFC/260 : SUCCESS in 1h 17m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"1e16a21487159d648504b2fb350b12322e8232b4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-18 02:56:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_Unity_FC/c203174/DellEMC_Unity_FC/2614 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_Unity_iSCSI/1d80016/DellEMC_Unity_iSCSI/2623 : SUCCESS in 1h 59m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"524b924df3ed7a0c0951ee07c973e7b72aa8e2ad","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-18 05:00:10.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/01/929701/4/fc : SUCCESS in 0h 42m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/01/929701/4/iscsi : SUCCESS in 0h 59m 01s","accounts_in_message":[],"_revision_number":4},{"id":"bb7b060bea88535538bb759a55decff7780064ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 06:13:15.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/571/","accounts_in_message":[],"_revision_number":4},{"id":"82e683762884724480254a8f73b4b7affa95f214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 06:13:18.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"23b24ed97846d6a0b740c93e0a613e04c8ba6b61","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-18 06:23:35.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"b306f140c70421880a81699fd8ba8ed897c9d394","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-18 11:25:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerFlex_3.6/b5873e0/DellEMC_PowerFlex_3.6/2108 : SUCCESS in 2h 23m 31s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerFlex_4.x/aa16f5d/DellEMC_PowerFlex_4.x/992 : FAILURE in 2h 33m 27s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/01/929701/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/e778818/DellEMC_PowerFlex_4.x_NVMeTCP/933 : FAILURE in 5h 52m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f7311e57b242854bc0c25f92a2f92a1f67054156","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-18 15:19:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/4/20240918201715/NEC-V-ISCSI : SUCCESS in 3h 48m 10s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/4/20240918201715/NEC-V-FC : SUCCESS in 4h 02m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e083f3fa8859376188d89ed28d164353feb703d2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 06:35:44.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/4/20240919065440/NEC-ISCSI : SUCCESS in 5h 15m 49s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/4/20240919065440/NEC-FC : FAILURE in 8h 41m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d7961671e7669ccf4a3f51cee913d3ef770c5a84","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-26 10:02:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9eb45b12b99562052cc593a952636b01babba648","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-09-26 15:35:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b60ac4e697621689b17b81c9358a7e2679a86159","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-09-27 07:01:51.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6fb85553fc088231c8241175d57cd318e5bf4402","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-06 15:07:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/929701-3/HPMSA-iSCSI/ : SUCCESS in 3:17:07\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/929701-3/HPMSA-FC/ : FAILURE in 44:12.28","accounts_in_message":[],"_revision_number":3},{"id":"db1e9d895b7c1701528160c9d30eed1d896907fd","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-10-06 23:45:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8086de7af76772ae391af593df80f354f632c179","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-10-07 05:25:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/929701-4/HPMSA-iSCSI/ : FAILURE in 3:28:22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/929701-4/HPMSA-FC/ : FAILURE in 44:32.58","accounts_in_message":[],"_revision_number":4},{"id":"31c72b765a048acdc4fd71678d00ce0286ac9a43","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-11 18:14:57.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"bbfb12a506f4a2cb3127a4eb0afd023d17403fcf","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-12 08:19:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/929701-4/Lenovo-iSCSI/ : SUCCESS in 2:23:39\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/929701-4/Lenovo-FC/ : FAILURE in 6:19:41","accounts_in_message":[],"_revision_number":4},{"id":"96409d8a3085aa83c2b07a07cab3f4e561f5cc15","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-12 18:18:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/929701-3/Seagate-iSCSI/ : SUCCESS in 2:24:54\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/929701-3/Seagate-FC/ : FAILURE in 6:19:18","accounts_in_message":[],"_revision_number":3},{"id":"cec4840066e912bd92bac1006a4fcb73f6ca8943","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-13 12:20:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/929701-4/Seagate-iSCSI/ : SUCCESS in 2:21:29\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/929701-4/Seagate-FC/ : FAILURE in 6:18:49","accounts_in_message":[],"_revision_number":4},{"id":"d8f99993506e8da0ed22032ff14d8e916c515bf7","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-10-29 16:14:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6345b50f0a36e35304f52e7fc8b72ed237539171","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-11-26 05:32:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd5811f3920d65b7a35dcf2e2cd2dbdcd9a20c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 05:37:40.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0972fea0549a20a0ab15029b3cb709842d3457ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 05:48:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f416de66eb54caf908db7262ae2b97c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae18a6ca17145d3bd83fa098a756ec5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73cc9421313e4ba4aaf15b75236e1431 : SUCCESS in 10m 31s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d072af7a9142403fa8008f3144042e48 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"b395351b1c84200765fa6dd18053b34a5136c936","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-18 05:49:50.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"767b719fd23c0596a2ab3466d4116e166adb2206","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-18 05:49:56.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"b1c2b32668dfa12bb5acdfa94989f1c6e3902502","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-18 05:50:35.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"252040277226bcc916a0789a25f5d3bf42deea71","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-18 05:50:47.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"0962507a6c8d7b3d11a220cfcdd8d14345bebca7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-18 05:54:42.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_01_2025_05_48_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e891e5cb6757c4422777b7d8f8c9786f6669ae2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-18 07:05:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/01/929701/5/ : FAILURE in 1h 26m 12s","accounts_in_message":[],"_revision_number":5},{"id":"7bc15fe520331a98b2be0a420e01b70601e7b5f9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-18 07:06:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/929701 : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":5},{"id":"dc19b16554b97d34abbf3e606e7fe1abfdf51590","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-18 07:44:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/929701 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":5},{"id":"12a3860e675b8b89591a6e9f69a7f5beb12e3789","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-20 02:23:01.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/5/20250120100048/NEC-V-ISCSI : FAILURE in 47m 22s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/5/20250120100048/NEC-V-FC : FAILURE in 1h 22m 10s","accounts_in_message":[],"_revision_number":5},{"id":"dfa7864fb390adb8c0c8afefa16c31dfb6a3e704","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:14.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1236/","accounts_in_message":[],"_revision_number":5},{"id":"2ba921c33956ace8163561368269e722ba296bee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:16.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/929701/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"75cc1315b5968c6b69e6fb57e2cc45c387984c2d","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 08:53:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/929701 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":5},{"id":"42eaddbd87e2179e16cbf300edac3c2d481f2a2f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-17 09:25:23.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/5/20250217165741/NEC-ISCSI : FAILURE in 51m 11s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/929701/5/20250217165741/NEC-FC : FAILURE in 1h 27m 40s","accounts_in_message":[],"_revision_number":5},{"id":"321ac30a0c92a86f28452988200dd3267475be3c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-24 18:10:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d9653f7aa944a6a33f4fca72c69c4c654f480d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 18:10:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5277e1aed7aa1027b7a5903a3b0f77c8f233ed69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 20:54:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a7b65d478094d2b9745714bf8534f25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161a35a2ba604ef2baf0d4c6a3b84b70 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18b80693a3b48b78938aa3042ef7bb0 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":5},{"id":"97c8aa55d56ca7ce8b75eacfb3bfac8385606438","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 20:54:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a70ecaaed9611772b441634493f990d3ea85f7a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 20:55:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fd866ed6a94ecf8895b2d797625629\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3742cb0b1674928a6fe7b38f9d75e83 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"12d91a893cf9498b91013e64ba44ed267cbe0b22","revisions":{"026a1288b0e2bc321fa11c72eb9bd10b8bd34a87":{"kind":"REWORK","_number":1,"created":"2024-09-17 19:40:18.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/01/929701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/929701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/929701/1"}}},"commit":{"parents":[{"commit":"7affd8a8b4e7400da16bcf37ef995bcc111d9bc3","subject":"Merge \"Hitachi: Support for QoS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7affd8a8b4e7400da16bcf37ef995bcc111d9bc3"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"subject":"Updating Lightbits\u0027s Cinder driver documentations, renamed mentioned LightOS with the actual product name Lightbits.","message":"Updating Lightbits\u0027s Cinder driver documentations, renamed mentioned LightOS with the actual product name Lightbits.\n\nChange-Id: I116dc49ebf2323480278c308ec2635fc866b5d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026a1288b0e2bc321fa11c72eb9bd10b8bd34a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026a1288b0e2bc321fa11c72eb9bd10b8bd34a87"}]},"branch":"refs/heads/master"},"b6237a1825d9cbb9a6690959989160b755ac75fe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-17 19:42:43.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/01/929701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/929701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/929701/2"}}},"commit":{"parents":[{"commit":"7affd8a8b4e7400da16bcf37ef995bcc111d9bc3","subject":"Merge \"Hitachi: Support for QoS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7affd8a8b4e7400da16bcf37ef995bcc111d9bc3"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:42:43.000000000","tz":0},"subject":"Updating Lightbits\u0027s Cinder driver documentations.","message":"Updating Lightbits\u0027s Cinder driver documentations.\n\nRenamed mentioned LightOS with the actual product name Lightbits.\n\nChange-Id: I116dc49ebf2323480278c308ec2635fc866b5d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6237a1825d9cbb9a6690959989160b755ac75fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6237a1825d9cbb9a6690959989160b755ac75fe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fa125abb0eba188bcb853fdac722484105d5e93c":{"kind":"REWORK","_number":3,"created":"2024-09-17 20:58:31.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/01/929701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/929701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/929701/3"}}},"commit":{"parents":[{"commit":"7affd8a8b4e7400da16bcf37ef995bcc111d9bc3","subject":"Merge \"Hitachi: Support for QoS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7affd8a8b4e7400da16bcf37ef995bcc111d9bc3"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 20:56:32.000000000","tz":-420},"subject":"Updating Lightbits\u0027s Cinder driver documentations","message":"Updating Lightbits\u0027s Cinder driver documentations\n\nRenamed mentioned LightOS with the actual product name Lightbits.\nfixed text indentations.\n\nChange-Id: I116dc49ebf2323480278c308ec2635fc866b5d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa125abb0eba188bcb853fdac722484105d5e93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa125abb0eba188bcb853fdac722484105d5e93c"}]},"branch":"refs/heads/master"},"962c35acaf2e258a154da549c89d6f9659e5d8d8":{"kind":"REWORK","_number":4,"created":"2024-09-17 22:54:56.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/01/929701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/929701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/929701/4"}}},"commit":{"parents":[{"commit":"7affd8a8b4e7400da16bcf37ef995bcc111d9bc3","subject":"Merge \"Hitachi: Support for QoS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7affd8a8b4e7400da16bcf37ef995bcc111d9bc3"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 22:52:00.000000000","tz":-420},"subject":"Updating Lightbits\u0027s Cinder driver documentations","message":"Updating Lightbits\u0027s Cinder driver documentations\n\nRenamed mentioned LightOS with the actual product name Lightbits.\nfixed text indentations, also added a line for paragraph\nseparation.\n\nChange-Id: I116dc49ebf2323480278c308ec2635fc866b5d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962c35acaf2e258a154da549c89d6f9659e5d8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962c35acaf2e258a154da549c89d6f9659e5d8d8"}]},"branch":"refs/heads/master"},"12d91a893cf9498b91013e64ba44ed267cbe0b22":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-18 05:37:40.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/01/929701/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/929701/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/929701/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/929701/5"}}},"commit":{"parents":[{"commit":"7affd8a8b4e7400da16bcf37ef995bcc111d9bc3","subject":"Merge \"Hitachi: Support for QoS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7affd8a8b4e7400da16bcf37ef995bcc111d9bc3"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2024-09-17 19:37:25.000000000","tz":-420},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:37:40.000000000","tz":0},"subject":"Updating the Lightbits Cinder driver documentation regarding renaming.","message":"Updating the Lightbits Cinder driver documentation regarding renaming.\n\nRenamed mentioned LightOS with the actual product name Lightbits.\nfixed text indentations, also added a line for paragraph\nseparation.\n\nChange-Id: I116dc49ebf2323480278c308ec2635fc866b5d7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12d91a893cf9498b91013e64ba44ed267cbe0b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12d91a893cf9498b91013e64ba44ed267cbe0b22"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
