)]}'
{"id":"openstack%2Fcinder~930314","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I63efc0598d501d77474588a02582f5338bb8d345","project":"openstack/cinder","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-09-26 23:14:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63efc0598d501d77474588a02582f5338bb8d345","subject":"Make default-types APIs compatible with V3.67","status":"MERGED","created":"2024-09-24 13:08:31.000000000","updated":"2024-09-26 23:16:10.000000000","submitted":"2024-09-26 23:14:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930314","cherry_pick_of_change":920308,"cherry_pick_of_patch_set":1,"meta_rev_id":"f77a5301fb5a4e2c024aa92d63d1a38219b1508d","_number":930314,"virtual_id_number":930314,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 23:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-09-24 13:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-09-24 14:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-09-24 14:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-24 13:08:55.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-09-24 13:19:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-09-24 13:35:43.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-24 13:55:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-09-24 14:10:55.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-09-24 14:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-24 14:26:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-24 14:27:47.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-09-24 14:52:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-09-24 15:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-26 21:51:36.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"028818c360904142c3a3066e5dd5b5ca1b92f3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-24 13:08:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"04a1fce20e59d9ebf558c271aa389d0a1ed875ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-24 13:08:55.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/617/","accounts_in_message":[],"_revision_number":1},{"id":"a65d7f028de1bf55b844d1f63c97f2f02bd95395","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-24 13:14:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/930314/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/930314/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"968433b31f7a91bba4ea260062f146d5e1e6e182","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-24 13:19:23.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e842f8b894d2400aba8631b16db569c4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a53db61466ae4ed893e1498e70423925 : FAILURE in 6m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc5df66a6bb0458a9b445a5cc322a506 : FAILURE in 6m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c09583503f124874a006718d3753d2d8 : FAILURE in 6m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f6cc6402e704cd997610893591f4748 : FAILURE in 6m 51s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devst... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"c10ca1be70452e5897932e492802d53f2515c109","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-24 13:35:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/930314 : FAILURE in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4a5f58fa09c34addf51d1b06e390bcd0ad6f2a42","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-24 13:55:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24f3cea9b442b9ed5a3a4ee2416493a372a648eb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-24 14:10:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/930314 : FAILURE in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d905a292364110787c8f144079294d20d8ce643c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 14:24:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99f80e8f3aa49aa8add42ab512d28ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e32487d130314ec2aa70f8a0b7e77076 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5657aa9e3d7e4c5aaa14d0cb54a45d4d : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d29c66083a881f947ea7c0b2492ff94184146212","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-24 14:26:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/930314 : FAILURE in 1h 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c272b0953d02c20f81e145fd26ab5d430f87cbb9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-09-24 14:27:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4adb9fec420b8646a3cf19b9777f587e673243ce","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-24 14:52:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1173/.","accounts_in_message":[],"_revision_number":1},{"id":"22ed73160170f226cc93aa845fadfbadc4d5b5b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-09-24 15:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/585/.","accounts_in_message":[],"_revision_number":1},{"id":"e99866262d216df4b31b26bc6a2d95316b4df455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 15:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08c17ef3701645f49b276d3de1370e83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84dfcbc78f24f0887dec0fd37633861 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee56043334024c44b5232909be61861a : SUCCESS in 10m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46e32df8cdb64e2d986130f4b0d46da7 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e820ea0016f4d65b86230ccf2037004 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a414ed50710c46cb9ef23455bb713b61 : SUCCESS in 44m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/de447be8223e423fae2e90a673cddf85 : SUCCESS in 52m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0ee487ebb044ebaab7c7a7913211008 : SUCCESS in 1h 29m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bda3a1d2430a42fd82d28c892c034a9c : SUCCESS in 33m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4550665525a844a98afa11a253738cba : SUCCESS in 17m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/991149e29ae2406f9e62d2783647af83 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2460d7b2915748c6853e11f916050fe4 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3262dc7cb57b4b41b122a5ef49904e61 : SUCCESS in 4m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e771f0dddea741e4bf41bdc0b4f863d7 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9596039a9724e0eac3bab47f9f013dd : SUCCESS in 47m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f54d5a733f4479aa9d631e8ea3321fc : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16779aff733247bead913543278f3fa9 : SUCCESS in 1h 46m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cbc44a4c27ee43ef8efeb2f903fe02b5 : FAILURE in 1h 54m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/136cfa9bbc784b61922770747f3ee6f0 : FAILURE in 2h 00m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/85fd97c39615486bafef072e969dd501 : SUCCESS in 1h 53m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ac4e101c46d4145924f48f79110da05 : SUCCESS in 1h 22m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb21791198c748faafcf5954d21d56af : SUCCESS in 1h 31m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c303b14587314454a139b8a6f9cd4541 : SUCCESS in 29m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a263feaebcab4cc78e092651fd5c4df2 : SUCCESS in 1h 07m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c0765ab5f39749d1bf5b1cfcd2a8d0bc : SUCCESS in 1h 13m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e54463e665d44daac0d748c72f219dd : SUCCESS in 59m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c16a18c8d848aaa6f68cc4a8c16292 : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":1},{"id":"378af4a3916385e0dffc2869d1a8cebd86fc1680","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-25 13:42:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ddd76e78317d439a4dcd8d08d1e3856a51976e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 15:22:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/697d0ceae87f4eac82d7badeb49b56f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8554bf91352b4670af9d112e692553b0 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c8feb909de4429eb2154ad9c4c16fb5 : SUCCESS in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd698f6c1aac4b8715eac65bf6d2e0e8f6dff735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 16:20:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78cde23ce18b4193b436c3f641aadb9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d02f281cdcd4e8690f9a89a4178b7b9 : SUCCESS in 10m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c8ab4de97f4dec891fec2bffa1dc53 : SUCCESS in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e9dbdd83cb046abab74284ce7d1f674 : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9729a914ac81477ca637a98919c45193 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/71d424679f5c4746b143d57aa978a16a : SUCCESS in 1h 17m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/97e075f3a408459da0a9fc8d62e60f68 : SUCCESS in 1h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f647168cbab4e8b9b6ca2c76dcca11c : SUCCESS in 1h 31m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/defc0c41bf0b4b988489e8cbccae2738 : SUCCESS in 19m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e76d9ae83fa4b2c8d3f671763d8b2c2 : SUCCESS in 17m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/edc2bb60fb5d43499172ef71b7bfda42 : SUCCESS in 3m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e03aa4f2daf145f59cad0ecfbe96e21d : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21fa2960e167432da3b6dfbad62e5193 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97a799d3e6d444af83821f4a08a9bf7f : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02457495fffb4766abc8e1003fb8f714 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83aef61ca0724416a63488b190553db4 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/577d9d8eb8da4b89a425bc526941c504 : SUCCESS in 1h 36m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9bb47c71bad241e09eb17752cf9301d0 : FAILURE in 2h 24m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91c7d635c17f419a86761d0820966154 : FAILURE in 2h 06m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/606a82b87de640c2b174472666da58b1 : SUCCESS in 1h 47m 49s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00d9cfdb45f64728a4bbc62618b5da21 : SUCCESS in 1h 20m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/019644d6575b477caf3e54b69606c0cc : SUCCESS in 1h 21m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/941d5b9315a4483faed40bc5e5e7332a : SUCCESS in 44m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a092f1da7694b4685b54a7f5f874e47 : SUCCESS in 1h 33m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c829aa8c47a2446c980d893daba3ddfd : SUCCESS in 1h 44m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3691f48f3b44854b0f23959baf06f98 : SUCCESS in 1h 35m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d48a5dabd344ae4b4c91a376f17fcfb : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7863a27ee679fcb371056b20d787d313be109407","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-26 08:38:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cc329e21af5889f016f28e13e7e15e80ea45c9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 10:50:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6d6d7b068e41258138edcc3426ac88\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/279400a115f644d6b9cd84082d43d66f : FAILURE in 27m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3398d306ed6f4315a896e947c635a6ac : FAILURE in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07a940d6d0402373150740d30b52a49c1fb3bc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 11:19:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0321421a9747b89cd66118f5de4516\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b791e2713f974007a210d25a19745c73 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d96d6743b348babc00156d1dadaaa7 : SUCCESS in 10m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65c061c327ba407bb15fa7e2b762aa1a : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fabdad80d114a80b551c4f8ced7035c : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/36a910c94aaf4bd1b85d07c01fa1d963 : SUCCESS in 1h 05m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/96c31ae2a2304a82877cba9360cb00e8 : SUCCESS in 35m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7af3ea64decc424793de1676d7dccb47 : SUCCESS in 1h 25m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05bf649dbe36443c8c35ff34317358fa : SUCCESS in 30m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84798ea74fcf4a2a9f161b0a76ba25df : SUCCESS in 16m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ae4bf7211c64774bbd8546d635223ae : SUCCESS in 3m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/74f396255cb34369be9d0757328191ac : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18abf2cfd65349a7a3a7987d00d35361 : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1bca5668a744433c8173ee7eb37c2e25 : SUCCESS in 4m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50e93f4e48574165ab30618722b7dfc3 : SUCCESS in 39m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54ac8c8b88c744dc84c3712a3b6cb026 : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6fcb629cb57c4679860cdaf25d29b71d : SUCCESS in 1h 04m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/59c9c9066bdb45d1868ef1b36fdf1f7c : SUCCESS in 2h 33m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1f775579bc1455abf30c61ebd3dc045 : FAILURE in 1h 49m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4054b083c20640638fd057f961b70be5 : SUCCESS in 1h 49m 02s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/803215ae6367449baf1a83f768e42679 : SUCCESS in 1h 01m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58b822059ba64afab153d80c02bf006d : SUCCESS in 1h 24m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef3e2c57ed32471c92b763f8a7417edb : SUCCESS in 33m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7d9ab9428da24814af04edecc0382520 : SUCCESS in 1h 13m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/24a9622d18f7408da3266968f3b40719 : SUCCESS in 1h 30m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/597db44298dd4dcab682bd393529133d : SUCCESS in 1h 28m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0341801ba5e34ae9b00fdb515661f692 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b7f92023adf1af8d80458199fda3530ab7419bd1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-26 19:20:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e983f131b3e43da36863fef8d45e5729c13b05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:32:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd59eab9c364bc68f0d7334238c677c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81332a3de3314b4185181c374c4b3a01 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0461c81a166347818f6dab4910504f94 : SUCCESS in 10m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c386838801e24687b6ee95225312f4cb : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85106182cb994cd4b5050985d9fc3d11 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/3f67fc42ff9a41b9ab4e609f5743777f : SUCCESS in 1h 04m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/80c0ecac70b540d8a257ddafdf3e22f5 : SUCCESS in 1h 01m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2650fbb3eab4878a28254a2e58b0bcd : SUCCESS in 1h 22m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a447411aa3647c69fc0bddfe28e882f : SUCCESS in 27m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/69cbe4c3c20941a0849c3d6da7b2c4d7 : SUCCESS in 16m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c766633e03f74a6fb3cd929597b9b6b6 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7261f5c308c428198bb990dcb4459df : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a91f9ec264447688ecac45da26fb443 : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/42bf45647a4e4b3f8e1e0fe0afda6a28 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/918edfe08ed346e1b2672774240b68d5 : SUCCESS in 31m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6248aa1a0c8145f997644e5d921615f2 : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/588eb4dd26ae474c8b13219e8589cda6 : SUCCESS in 1h 29m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d6c77c82efb74433a5c8d5cb867d689d : SUCCESS in 1h 41m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/966368ab9817481aba283a42100a3bbc : SUCCESS in 2h 03m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c21ec53fa0d140d781cf99e4125e24b0 : SUCCESS in 1h 28m 08s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/779d7162485d43d89631a2e02df8abd4 : SUCCESS in 1h 11m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7df96c661b754de2b3cddceabe18dd8c : FAILURE in 1h 15m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8cbbd8b8b4b24709bbc64fdcfe144b42 : SUCCESS in 32m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/94ea56fdc6924e8fa71eb53d4db3181c : SUCCESS in 1h 12m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7a8df7fd584b4143afaf210c086f4c75 : SUCCESS in 1h 25m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93487496a3fb4526a45d9df5e130a41e : SUCCESS in 1h 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afb477ca6f4449f9964e4d78e1eee9cf : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b1e194bab0d8b0db889b5417f35cf20eed2157f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 21:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6557a8be5bd6f163f31dce0d08fd933d48babc16","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-26 21:51:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/930314/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f615c20/ : RETRY_LIMIT in 7m 49s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/14/930314/1/upstream-check/cinder-tempest-plugin-ontap-nfs/f3bcf6a/ : RETRY_LIMIT in 7m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ef39cf913c02d068d97a3f241b151ebfec1ef2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 22:05:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a46b0fa404e4780936d1900b3f7d48f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74341559e1114c2dbc482e4876700707 : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0284b853f5af47328c555741cec745fe : SUCCESS in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a6d6b4005b845a7f325357f9430deedb4b3b985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 23:14:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781a807c9ec948c8baa9103fe50395c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2d9cbaa9f445abaf3b0e50ccd50ac7 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b9a2e6bd12146cfb6b8ffe6b3f6b0f7 : SUCCESS in 11m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35945ab713144457a8731e2ff3b9e072 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e089b1345cb042fd8df7d4e70e5001fe : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/ab5949132b5c43e692be9e32831b0252 : SUCCESS in 58m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/11ae6e151f4f4ef691c91a628e3ca505 : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f315470621a46ad9778e0ff0e546ee2 : SUCCESS in 1h 41m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0eebad4c12b457dbd02bb5f57037f0b : SUCCESS in 20m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cebd1c5cde34a55b9700379631dbf5c : SUCCESS in 1h 11m 55s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ec580b9fb65a4243b5be3d7cb056a173 : SUCCESS in 1h 12m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc87a4c8e116439684bef3ae99e6da16 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"53d98ade9e0e565e54a8563254cc42566d59de5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 23:14:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f77a5301fb5a4e2c024aa92d63d1a38219b1508d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 23:16:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2790a16084491b9fa30ae8156aa374\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec1bd32c8a044cd8caa8d56f150fbba : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bb2f13de33d509967560d5df0b47c32bdd17007","revisions":{"2bb2f13de33d509967560d5df0b47c32bdd17007":{"kind":"REWORK","_number":1,"created":"2024-09-24 13:08:31.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/14/930314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/930314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/930314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/930314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/930314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/930314/1"}}},"commit":{"parents":[{"commit":"530376bc5d9bb7a4a6cd70a723f3b3d4a7d309c4","subject":"Add cinder-manage command to update service_uuid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/530376bc5d9bb7a4a6cd70a723f3b3d4a7d309c4"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-05-23 17:26:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-09-24 13:08:31.000000000","tz":0},"subject":"Make default-types APIs compatible with V3.67","message":"Make default-types APIs compatible with V3.67\n\nIn microversion 3.67, we made project_id optional in the URL\nfor all Cinder APIs.\nHowever, the default-types APIs (set, unset, get, list) were\nimplemented without the project_id in the URL to make it\nready for SRBAC and anticipating the new endpoint for cinder\nwithout project_id.\n\nThis is causing issues while implementing the SDK support for\ndefault-types APIs since we fetch the endpoint containing\nproject_id (http://127.0.0.1/volume/v3/\u003cproject_id\u003e/default-types)\nbut the default-types API doesn\u0027t expect it\n(http://127.0.0.1/volume/v3/default-types) resulting in 404\nresource not found error.\n\nResourceNotFound: 404: Client Error for url:\nhttp://127.0.0.1/volume/v3/default_types/12a1b5e507e7497db79707b0ddedf1a4,\n: 404 Not Found: The resource could not be found.\n\nThis patch makes the default-types APIs consistent with the\nother cinder APIs that accept URL with/without project_id.\n\nNOTE: This patch doesn\u0027t require a microversion bump or releasenote\nsince the expectation with MV3.67 is that all APIs support\nrequest irrespective of project_id in the URL and this change is\nmaking default-types APIs consisitent with MV3.67.\nAlso MV3.67 is just an indication of the behavior change otherwise\nsupplying URLs without project ID with MV \u003c 3.67 also works.\n\nChange-Id: I63efc0598d501d77474588a02582f5338bb8d345\n(cherry picked from commit 9afa19e9c96831b319c53fd222d7b822e52da967)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bb2f13de33d509967560d5df0b47c32bdd17007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bb2f13de33d509967560d5df0b47c32bdd17007"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
