)]}'
{"id":"openstack%2Fcinder~931510","triplet_id":"openstack%2Fcinder~stable%2F2024.2~Ifaceb04263973b1445c7f13a6155dc63647d7176","project":"openstack/cinder","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-12-16 22:16:34.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-12-16 22:16:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifaceb04263973b1445c7f13a6155dc63647d7176","subject":"RBD: Fix handling of RBD errors in get_manageable_volumes","status":"MERGED","created":"2024-10-04 16:42:15.000000000","updated":"2024-12-16 22:19:26.000000000","submitted":"2024-12-16 22:16:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931510","cherry_pick_of_change":919607,"cherry_pick_of_patch_set":5,"meta_rev_id":"123ffe397be93a7c2cc179e926283860f25fda06","_number":931510,"virtual_id_number":931510,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-16 22:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-12-16 20:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-10-12 14:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-24 14:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-12-16 20:02:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 16:42:41.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-04 16:46:14.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-10-04 17:09:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-04 17:14:18.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-10-04 18:18:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-04 18:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-04 18:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 18:57:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-10-04 19:11:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-12 14:17:51.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-10-24 14:24:12.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-12-11 16:02:07.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-12-16 20:02:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"5c2905972fa6e0e1e87cfa1fae44543adb1b3321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-10-04 16:42:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6116ea0ddefd43459ef42bf2dfbb9f4a816a4bc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-04 16:42:41.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/686/","accounts_in_message":[],"_revision_number":1},{"id":"e9a8bea46acce2a1e88e44b8580956c911fefefe","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-04 16:46:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9f21e40009ec6b94d129fa38528d9f5ebabf144c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-04 16:48:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931510/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931510/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"9ed71103a8f07b28d9cc64cc4a6466977ae7319b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-04 17:09:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/931510 : FAILURE in 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d7180dfeb334ec9845719e03b3734510ed1ad983","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-04 17:14:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/931510 : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"27815b27726f4c48abe753d93ce693c2ec088220","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-04 18:18:42.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cc8f759f8e9341258664e0b50690ae32\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3aafd466fb6d41078d8d0e1f15902e1e : FAILURE in 1h 12m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e817485398964957bac7651104484f99 : FAILURE in 1h 25m 31s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83fa193648bb4ddeb7630aca0c9a3bd2 : FAILURE in 1h 35m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0b4f352158e4cc892c5bad7ac21705a : SUCCESS in 35m 32s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"adc6e02590263968529e50dc374b2f558cf5edef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 18:51:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3dfaa3964884a9bab4864081944de8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab7e4c162c3949db9ca1aa48466207a5 : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee8b6f1f7fc542b298aef34e79b6c538 : SUCCESS in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8025399903c3de6dc194f563c67f4c0eaf20135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 18:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/714f52e791714693bd3c2e6b4e0f3fd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4654fd6952a6485b98741a1a5a1b501e : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b2cd6e0638c40d29ff6a5eefe5e7b87 : SUCCESS in 12m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51af2561757441d7a9b08877abf4e8ed : SUCCESS in 11m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfca5ce4e5f34db4817302ca0017ef7e : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97665e34f3b4511aa794e55debf9d87 : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee3bd674ea0497cbc4e08d9b487655e : SUCCESS in 54m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2752a0841c0940e7be3b48ad79194282 : SUCCESS in 1h 28m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eda35bca1edd4f19b31b2720fe388dfb : SUCCESS in 1h 05m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fed3c6d405646aeabccf697e73830b9 : SUCCESS in 19m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3faae4c58a1a460db509a2bba4293572 : SUCCESS in 17m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/69928ec8042f42eaa1c0537732333d52 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba8e7f474e324c9884e222bece2d8050 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35cf762097024aef93a0f909c5444a46 : SUCCESS in 5m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0f635af3e9a64b78a34269ad512cca36 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/34d39c812b914833b2244e59d81a75ff : FAILURE in 23m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30386ce8b5fd4a1280729951b086c61f : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e75e639db0744dd4ac29c2789c5c075b : SUCCESS in 2h 03m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3752a724486433aae026af7375f3654 : FAILURE in 1h 48m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/be9d8707c5864ff3bff539cd44ba718e : SUCCESS in 1h 51m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4b2bc7288da9445db30adf9547562982 : FAILURE in 20m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6d9a95d62e5441239964059a0c8537e2 : SUCCESS in 38m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4fee79ad212c4d629654e48020961eb2 : SUCCESS in 1h 09m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9c4ea3c750824a709ad195c1a5480367 : SUCCESS in 1h 25m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc02f190dc6c4aaeb0e43ba9318c70f5 : SUCCESS in 30m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/47ee9ac9b468473c88364e5b7628129b : SUCCESS in 1h 24m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cb545b74408d464db431d30fbafc253f : FAILURE in 8m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51c7f1fa5efa4e5aa14999dc5a582632 : SUCCESS in 1h 22m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea8056a69ab5475a8795b85c7b1f1923 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b206a3279275b0a8a65534257ce26b329d1392c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-04 18:57:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_10_2024_18_52_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7e06010702ca03b56d17895ffe916f439ba2d0f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-04 19:11:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/931510/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/407a06b/ : RETRY_LIMIT in 7m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/931510/1/upstream-check/cinder-tempest-plugin-ontap-nfs/0a3da7f/ : RETRY_LIMIT in 8m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b5b1aa770cdfefd2bf43c9871b23bb5dea207e3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 14:17:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e8c5453beb1449d890b7b2d6a9da7fa5c482c81","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-10-24 14:24:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e3c6a23a5925ff0bd97afbfd7090b50323aeeff","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-12-16 20:02:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5efc7c2efb07a382661fe481da84de55bc90b097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 20:03:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bd5305d1db1245f7396eb8a8c292fa758a322b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:16:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cac1ea4dc6045d7801bfeb9a91722d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdeedf8d30ea44398c91950d0c9fb7af : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deee5212c6934ad0a79eedc684534cca : SUCCESS in 12m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b32bfb7abe3a4d12a87726c552c86176 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590edc2cd32244ffa75170f69bcd9689 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/d202572704ad4d029025bf837bbd0668 : SUCCESS in 1h 14m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/097c3fda70eb41d188946a3bc67ad8af : SUCCESS in 1h 35m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f128143fbf2d47499a5bfe0b0d976a35 : SUCCESS in 36m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22fd3e16ddb14c809fd9cdeb5d153ed3 : SUCCESS in 17m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3feb5519f20b4959b0d0470a84c20348 : SUCCESS in 56m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0f44d1c64a54091b33aade8838b766f : SUCCESS in 2h 03m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9f22be0fa10449487d18e60a553759d : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5d9e53f43863b14ada7d19fbf76120f6a588c916","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:16:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"123ffe397be93a7c2cc179e926283860f25fda06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/253853f2c7dd4df4800c768bf3721374\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770fc3e541fc420fb73d9b9e4a0a6e59 : SUCCESS in 2m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad1c0b41b80e455295a180ed7e9a0e16 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42ae198f19962ff8d847fe8995cf750ec2830290","revisions":{"42ae198f19962ff8d847fe8995cf750ec2830290":{"kind":"REWORK","_number":1,"created":"2024-10-04 16:42:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/10/931510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/931510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/931510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/931510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/931510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/931510/1"}}},"commit":{"parents":[{"commit":"6ba70ec68493eaf7c17eea05aa33b4086423e0cd","subject":"Add cinder-manage command to update service_uuid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ba70ec68493eaf7c17eea05aa33b4086423e0cd"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-05-14 17:33:48.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-10-04 16:42:15.000000000","tz":0},"subject":"RBD: Fix handling of RBD errors in get_manageable_volumes","message":"RBD: Fix handling of RBD errors in get_manageable_volumes\n\nThe \"except ImageNotFound\" was outside of the required try block,\nso ImageNotFound errors would result in exceptions being thrown\nup to the API layer.\n\nAdd handling for other errors that could be thrown by RBD\nhere, too.\n\nReference info:\nhttps://github.com/ceph/ceph/blob/08d7ff952d/src/pybind/rbd/rbd.pyx#L8\n\nCloses-Bug: #2065713\nChange-Id: Ifaceb04263973b1445c7f13a6155dc63647d7176\n(cherry picked from commit 7f49b41eefbbbe5b5b7f95b5d936ee9118861ba5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42ae198f19962ff8d847fe8995cf750ec2830290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42ae198f19962ff8d847fe8995cf750ec2830290"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
