)]}'
{"id":"openstack%2Fcinder~931888","triplet_id":"openstack%2Fcinder~master~I9e8e5f38c8ab8804c5e7ae4ad5e0f650768e93c3","project":"openstack/cinder","branch":"master","topic":"bug/2081742","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2025-03-01 21:08:22.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-02-27 22:03:00.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2025-03-01 21:08:22.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-03-01 21:08:22.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2025-03-01 21:08:22.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-02-28 09:55:24.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I9e8e5f38c8ab8804c5e7ae4ad5e0f650768e93c3","subject":"Dell PowerMax: Added exception handling after the masking view REST call.","status":"MERGED","created":"2024-10-09 08:46:40.000000000","updated":"2025-03-11 15:46:53.000000000","submitted":"2025-03-01 21:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"931888-bug/2081742","meta_rev_id":"14e5e61e44c4edec184aec5ef992061a8e4482cc","_number":931888,"virtual_id_number":931888,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 21:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-02-27 22:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-28 09:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-28 09:55:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37269,"deleted":true},{"_account_id":37368,"deleted":true}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 08:46:49.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-09 08:46:54.000000000","updated_by":{"_account_id":37269,"deleted":true},"reviewer":{"_account_id":37269,"deleted":true},"state":"CC"},{"updated":"2024-10-09 09:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 09:19:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 09:24:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-10-09 09:32:48.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-10-09 09:48:01.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 10:44:44.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-09 11:13:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 11:44:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 07:15:45.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2024-10-10 07:15:45.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-10-10 07:15:45.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-10-10 07:29:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-10-10 07:39:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-10 10:23:19.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-10-10 14:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-10-11 05:51:40.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-10-11 17:23:38.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-10-12 04:20:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-12 06:42:05.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-13 01:48:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-11-22 08:52:43.000000000","updated_by":{"_account_id":37368,"deleted":true},"reviewer":{"_account_id":37368,"deleted":true},"state":"CC"},{"updated":"2025-01-29 15:45:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-02-27 20:29:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-02-28 02:46:46.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-02-28 02:46:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-28 02:47:37.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-28 02:47:45.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"ff4629504d7f6b0cc342f55617870f3e5dc76418","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-10-09 08:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9af6cdbcabc57b127c1410f2d5c27f794e133cc3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 08:46:49.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/720/","accounts_in_message":[],"_revision_number":1},{"id":"aa1d01148daaad2371b36235050dcedb52975218","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-10-09 08:46:54.000000000","message":"Patch Set 1:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/45/","accounts_in_message":[],"_revision_number":1},{"id":"1a0cd0c1bfa4ae5bb99cf9c9d218fadf9d0e1783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 09:18:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/680e6c683f7748e995a87f798823654f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92b0aebb03c542faaece800b8eed8305 : FAILURE in 21m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f6b8aaf97e13463a930c9a85c90f4b7c : SUCCESS in 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75debe19051aeeed38f065cffc2f50fe61227535","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-09 09:19:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/931888 : FAILURE in 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a710403cdefc631ee780df5e0eb9402b1bf5b5c7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-09 09:24:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/931888 : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"53056efdae9c139eeed5c51bf4bdac4f78fbb0e6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-09 09:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/931888 : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":1},{"id":"39a3805cfdeb82c3814e9082d7032926c704ae51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 09:47:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"afbb43a77e920f10575f64fd3bc0c81691fab687","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-09 09:48:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/88/931888/1/check/fujitsu-eternus-dx-iscsi/ccefb77 : ABORTED in 1h 00m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/88/931888/1/check/fujitsu-eternus-dx-fc/10cf320 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3e268dd05dc045bcc0bfea1affebfc95db750a87","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-09 10:44:44.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2ca7ecd3f444429b9c60dae5fca423a1\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42e59bd9554e4eefae21d8eb4f14ab6b : SUCCESS in 37m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7dcbb07ec97479896384d9635d618d3 : FAILURE in 1h 04m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98e35f5fe0fa42bbbb73579e304ebb6b : FAILURE in 1h 57m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26d5294bf3c7474395306a5913ed0043 : SUCCESS in 36m 30s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03252de186364647a9fa4c99f31df8e7 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2eabb6fd20f14bc49c2195609f08526e : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e54bcd16755449d590d1ebf5bcc84346 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25f4f02b933d4e00a737e458e66d6a7a : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"c718b293c0d7e56be04f15be910e16b8e984230b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-09 11:13:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/320663d19dd6463e8496c1dc158c1959 : FAILURE in 1h 08m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6409412ae615413d8a917a5cbd7c0a89 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"395a2917f80ad1ad89e0f98239a5af093870cf0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 11:44:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccc4f8ff55c84fd8885ce61349b5b954\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59f407e8f7e436c9bc5c28fbda91f56 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c019645a78a4e959fe30eb5e9f5aade : FAILURE in 11m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e02cc3a0218a4d8aac34137e52e1c338 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ab486f24e146878478a808119e6cd5 : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/f7fc19197e4a4acb9498fa2f45c1a204 : SUCCESS in 1h 04m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad39a91b85e2415386451b0255759ff6 : SUCCESS in 1h 02m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a081e3a155794359a705ef710be8b8fd : FAILURE in 1h 22m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fa6a9c87a6e4a528856e3fef5ae8d37 : SUCCESS in 1h 03m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6cc19ac52414b74881adc43df7d72b8 : SUCCESS in 20m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f72d6b007d34fb9a3ab38d0c75616b9 : FAILURE in 14m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b3a0c986bfd24218b7a940d70d295efa : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2cdb1ede6eb4cce8131b286c0321e0f : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e59d96cd893439ab3565d40a292209d : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/17ac0e0c770d480293a96632304a67b4 : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ad3052093414a90b51b806bf1ec0695 : FAILURE in 26m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff5c3b5a3c6341928000c1375eabf82e : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a904654fcbaf40dd9417e73790f86f4e : SUCCESS in 1h 50m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a5c25a847ee747658b630018d3363e11 : FAILURE in 2h 20m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ecdd92004194a6dadaf2ab1345482ed : SUCCESS in 2h 13m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c0527d247fd94d918d384aa5c2a1f6eb : SUCCESS in 2h 00m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/525c86491b0f499a862891f9725733e1 : SUCCESS in 35m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a0cd55b843d4b8b9b34c0f3ec45ab86 : SUCCESS in 1h 19m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a3cc1a5588fa4e13b9133dd2ebd772fe : SUCCESS in 1h 28m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/97bdc421496d4444a60d3ca425b75585 : SUCCESS in 44m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf61d11098104f1aa70a8106bb1cbd8f : SUCCESS in 1h 23m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d32c738f4f941f4bd83d7651abb0b4a : SUCCESS in 1h 29m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/013e58dddc34482697d49e0c9cf75867 : SUCCESS in 1h 43m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84989785c76e4041b1807062051e1e53 : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":1},{"id":"34bf56e19bdeae7f9852220444d08a1c26094c21","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-10-09 11:46:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9f80274e641ab960367869c4ec4812485a9b05d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-09 13:15:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/93ae9d4e18f14318819edbc0c6ac8095 : FAILURE in 1h 19m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/71922555a7fe43e78c66f09cdaecd8ca : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8569fead1e96a1003e4a92eae775ee174e733664","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:43:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab1880c696374a258b495b88e5ff2594\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ab9358b794040fc9b53e2b9444cab7b : FAILURE in 30m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5da87b10458840b98cbc064fe4d35be9 : FAILURE in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fe37da36352835b9769c82884d8df73f4babb5d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-09 14:02:39.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e1c727916e64c6da6bd088691aa7757\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fda712af71954995b62a5f97a93b6c3e : SUCCESS in 39m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33128d3ac03146318dee27efe229bbdc : FAILURE in 43m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4feed84e48a7433abf4302ebe4dccf81 : FAILURE in 1h 40m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71c30141b66e459386c0bfbd6dc7e6f9 : SUCCESS in 39m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b061ba6854042a3928bf7df23dd6e8e : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5aabcef82e8f4784b86feeb2ea15ce97 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/376d0fa695854519bafd8652f4cd94f5 : RETRY_LIMIT in 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36cabd2040994974bfc28515fe5ee2d4 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"9f3de0f5ed0264eebf00fb636596b81719eace29","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-10-09 14:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/88/931888/1/check/fujitsu-eternus-dx-iscsi/2d552ff : SUCCESS in 2h 19m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/88/931888/1/check/fujitsu-eternus-dx-fc/875c90c : SUCCESS in 2h 08m 35s","accounts_in_message":[],"_revision_number":1},{"id":"236cac67221a4e18725eab4ec10590d4394a21ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:10:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/314b88b4053c47f8b39e402e51951f17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f2bca95ae64608b33fe1abb82d6a72 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31facee4d8a64a1aa5de6e358c3e3f2c : FAILURE in 10m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/006dec57dce045c79bd331c1ec0352db : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432ef38e844e4f8990a8f0a986e37e60 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7e795487a692459a8b56c657b395e5ac : SUCCESS in 48m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd7a914ff68f4828b1c22cb490ac8b46 : SUCCESS in 39m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/415dc333d3f04334b9cb05a2d8ecbc49 : SUCCESS in 1h 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37358f7fb8c64a62a0bed24a0f12cfdf : FAILURE in 57m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2591e4860d84ab5947096d338ff1aa3 : SUCCESS in 18m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5dd2618e82784fb084e4beedf3838538 : SUCCESS in 16m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/50c22ec09ab94158ac68cdb9a91747ba : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da60534c84944dd69c0524cd921f7c53 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de878ea7476a4f7ab5fe2b6be93a9aa4 : SUCCESS in 6m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f52874f6eb9c4e098960f80b7c67cb03 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9eccaed7122840c583c718fcb6e89785 : FAILURE in 36m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3286e6b852541c0b7a67d826eb25f5a : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fdae5af02c6540ec93cbf15f8f5e7ae7 : SUCCESS in 1h 59m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3953ec5027a64bd2a9f7bb17c90565f7 : FAILURE in 2h 02m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76852e800e684b6c8363ed9142c722e3 : SUCCESS in 1h 51m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7538eda9edfa47ef8af07e5536d51816 : SUCCESS in 1h 52m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/afaae373d098455b90cd9bf7e2c6ae6f : SUCCESS in 36m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1864fd379fb41fea13a4b33358c6089 : SUCCESS in 58m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/daa17f0a9ea84211a08a1745847da9d1 : SUCCESS in 1h 37m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0737cc51d5ab4cb18ed93e6701573dcd : SUCCESS in 41m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d375a2b45f6d4f3a965a121b3b2fe949 : SUCCESS in 1h 29m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ca4a23dd718b48b2904e2021f45034ac : SUCCESS in 1h 54m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52f153fe9f844d40835c77809f0efbe0 : SUCCESS in 1h 35m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2262005725c44d1b232621b069d2c1d : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a3d8d6ac88df8ae6170c6d322351f7a1db63b911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-10-10 05:15:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b9f0a1290bc365a5e81d76246e0f761d771d61e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-10 05:27:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/35fdab7be90e4e469553ad338f26342f : RETRY_LIMIT in 2m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9e04fcff3c5e44f99a961ff2bd857718 : RETRY_LIMIT in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0b64a8e3b223bc43d9a125e0dcfb7fed217b6cb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-10 05:32:05.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/724/","accounts_in_message":[],"_revision_number":2},{"id":"55a66c1d7b3336e76709a6f266cc583cf55daf04","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-10 05:59:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/931888 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":2},{"id":"fed06dca4c5af251f6832ead270aafef87ac21c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-10 06:08:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/931888 : FAILURE in 51m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d9c70151bac16b48ce94b9beccd028689b5ff8f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-10 06:31:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/931888 : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":2},{"id":"28c2c442d6da86701a52869438903243b01d14aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-10 06:31:46.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"2c74ef9b96fd896454a794b2cd9114a06ca41211","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-10 06:48:02.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e145e46a775a4cf39f2e65e13a44ba86\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9687ef5efd654548bddadcf76414ed2e : SUCCESS in 44m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/891edeb035af47c98932724cedcdb5d9 : FAILURE in 42m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03ac1ff526ec452b807fba3a97373c9c : FAILURE in 1h 32m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d21d6f3ee2248daa5aa4e6fd847bf78 : SUCCESS in 34m 14s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53db4f1e393c4885848be755cbd8ea30 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/597ff379aebb41c9b8ed836830e46acc : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5875f78c5ab44189a1542a562b442873 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7130a32c13984873a494b84144dc42a7 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"a849547d06f5006b126fdbbd3bc9269fe0f61921","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-10-10 07:15:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2fe0f84c4adf8e046803c71f05b12d3bf18e5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 07:22:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe9edddaaab43829a0dd2d1ea4782f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70df919a4a2c47268459516f5eceb123 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03556ecc6cfb4efbbf59886e11a4968e : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edb53764e1464b6eb6f7206d69a0e44c : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6723a504e84a7a8f20eced586f394d : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/830fbbb3f17d4dd79dd055562bde7601 : SUCCESS in 38m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/484275f16fac44f0881a776128ae9dee : SUCCESS in 42m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3e6691bffd741db9f1174a125cc97b3 : SUCCESS in 1h 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/526095d3e70049728886ec2dc873b992 : SUCCESS in 57m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/726d97246ab044788beeea54bc726285 : SUCCESS in 18m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1870a799d0214cb39168e93afc07afc6 : SUCCESS in 16m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e02f66bf71a4e179ee36eeb7af2293a : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f963479c15b44050829b64bc86531b9c : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0862f5da12941b8af3871481f808b0a : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/833573655a3c41a39f14867ac7d624fe : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eceecba0a54c4e21897703bf6bce6952 : FAILURE in 25m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac39831f49b24e6ba5204bfadb420195 : SUCCESS in 8m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2d34c8b50d84228a080d031328abc98 : SUCCESS in 1h 59m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/648d8868f93242dcbad30d4a9826e4d0 : SUCCESS in 1h 57m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ca22daf01f644d6b13833cac724b2da : SUCCESS in 1h 52m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e3a7945739154c669ca723e294bac10e : SUCCESS in 1h 28m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9acfb431336e4cddb7f43e0d3574ff73 : SUCCESS in 31m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e466cc56cb2490593df9eac287f24d1 : SUCCESS in 1h 16m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a420e84d24a4c76bcaebb2f561590cf : SUCCESS in 1h 45m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb4cb9439d6b42368e499f6aede40c19 : SUCCESS in 26m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9731fde50aa94ae281d57c9afadc361a : SUCCESS in 1h 28m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4130e3fc2e61442b8ad378ebfb69bb59 : SUCCESS in 1h 19m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7043ecf45e9f4d48b98ee8b6724f5223 : SUCCESS in 1h 50m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff395b8df6da4e73bb65835a176b385f : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c08e0775fad99fee755f44e971aded55036409ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-10 07:29:00.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/10_10_2024_07_23_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2328b981e9db1d6eb064ce904bcaf93b1ba06055","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-10 07:39:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/931888/2/upstream-check/cinder-tempest-plugin-ontap-nfs/cdedef9/ : RETRY_LIMIT in 5m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b2a9380128ab3320a967a98ea8b63f97aee004fc","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-10-10 08:09:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66cd7dfe78b7260555e065b748c29060a0129ef4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-10 10:23:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_iSCSI/b5b7fd3/DellEMC_PowerStore_iSCSI/1809 : SUCCESS in 1h 23m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_FC/9211035/DellEMC_PowerStore_FC/2007 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NFS/13b6241/DellEMC_PowerStore_NFS/1502 : SUCCESS in 1h 24m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NVMeTCP/fff2d4b/DellEMC_PowerStore_NVMeTCP/1361 : FAILURE in 2h 52m 49s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NVMeFC/a2653d0/DellEMC_PowerStore_NVMeFC/326 : SUCCESS in 1h 13m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5c3c67f8dfbe0a2fcf2d4ccce5d302ca27b574ec","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-10 14:00:10.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/931888/2/fc : SUCCESS in 0h 52m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/931888/2/iscsi : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5bc97f93a8483cc43676e9fc00961773a9d44db7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 14:00:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/153be35d9883423ca84c502c19590af2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5b788594a8f452489a9abfde66b99c7 : FAILURE in 22m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92ccf5083f1841a7acf803573ca9c038 : FAILURE in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e28eaf9d052a4b78b0cdaf86c6edf4a3c95e1dd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-10 14:17:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/931888/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/f8e1f52/ : RETRY_LIMIT in 5m 00s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f8ec7471106d09e7507e6a2bd17263b0d45a78c1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-10 16:46:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_iSCSI/493ad9a/DellEMC_PowerStore_iSCSI/1810 : SUCCESS in 1h 47m 59s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_FC/848abc2/DellEMC_PowerStore_FC/2008 : FAILURE in 2h 45m 53s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NFS/d0656d6/DellEMC_PowerStore_NFS/1503 : SUCCESS in 1h 07m 17s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NVMeTCP/509bf1a/DellEMC_PowerStore_NVMeTCP/1362 : SUCCESS in 2h 11m 47s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerStore_NVMeFC/fbf2d39/DellEMC_PowerStore_NVMeFC/327 : SUCCESS in 1h 13m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"14de631ca70ad50e893e0b5410528213b3411ed9","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-10-11 05:51:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6f04d356753ed45ffecc2cf9270220652ba5243e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-11 17:23:38.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/2/20241011175743/NEC-V-ISCSI : SUCCESS in 4h 04m 56s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/2/20241011175743/NEC-V-FC : FAILURE in 8h 25m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b21aa5415ee5ab333397154e4baa46da97ea5e32","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-12 04:20:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/2/20241012073730/NEC-ISCSI : SUCCESS in 5h 14m 04s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/2/20241012073730/NEC-FC : SUCCESS in 5h 42m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3cb95a06fdbbb1db54d048ce6aeffff73ebd8fa0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-12 06:42:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerFlex_3.6/4d53245/DellEMC_PowerFlex_3.6/2227 : ABORTED in 2h 16m 37s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerFlex_4.x/554daf5/DellEMC_PowerFlex_4.x/1109 : SUCCESS in 2h 13m 45s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/38d4a4b/DellEMC_PowerFlex_4.x_NVMeTCP/1035 : FAILURE in 5h 54m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f0825c44a5e7102a12c66a97e888bebf2a3eb688","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-13 01:48:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_Unity_FC/763c299/DellEMC_Unity_FC/2728 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/931888/2/check/DellEMC_Unity_iSCSI/5ed7a8e/DellEMC_Unity_iSCSI/2725 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e67a1f1c14fbb45c183c4246109b05c8d98f2922","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37368,"deleted":true},"date":"2024-11-22 08:52:43.000000000","message":"Patch Set 2:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/52/","accounts_in_message":[],"_revision_number":2},{"id":"7418e137a72ebe0aa9f0d839a10d7658de003b0a","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-12-19 04:37:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1555fb134b4385a740984d32517bc5411ad474a5","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-01-20 19:25:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c1c33153f3806f04b11182da806362b0b0e423b","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-01-29 15:45:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc7cf3ec0b5410bb369135fa0eed3b2633b1bf47","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-06 02:27:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9b12f665d208161559b666012ec6fbd58757f05a","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-06 05:42:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cdf86bf115ccf86f38b3f7d03e7bdcd95cd5f90","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-02-26 18:53:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"18dd0672f3da64c8fdb7aa054d4fe7f1d54bdd46","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-26 19:11:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1c6676a67f5765711f616ac241282e250707e95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-27 19:14:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d9ead28ccc9105f2d2e4a71b22a983b71ca1b8a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-27 19:14:26.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1379/","accounts_in_message":[],"_revision_number":3},{"id":"35ef6e8d3b8b3f409d9b890c1f4e80549bcf37ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-27 19:14:28.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931888/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"36c8e52e8c0df352bda9f8fbf4fee13bd8067127","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-27 19:17:14.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b39730eb9eba5ca0ae958386cbde4b4df5a8081","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-27 19:58:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/931888 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":3},{"id":"0668ccf27f6685a63bc2a82d67c6953f48131c97","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-27 20:29:43.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/408/.","accounts_in_message":[],"_revision_number":3},{"id":"76d9d7899f3a77813ba65ab14a2347e0677c8588","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-27 20:37:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/931888 : FAILURE in 1h 05m 44s","accounts_in_message":[],"_revision_number":3},{"id":"cce08e32ddc25be69f7ae5a448b2e6531d371d7b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-27 20:52:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/58a8a660364d451e8931c87d79a4cbe7 : SUCCESS in 1h 17m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f0225693b98a48959f249fd8e1724fee : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b310067b9f380194c659c05d393d4247851e19e9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-27 21:02:39.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ab01ed4945f841a68d043e443a357984\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69a3976e87aa48c4bb1dfaa8ae739376 : FAILURE in 53m 26s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f8c70a035a84cf8a5c43afacbfdb7ae : FAILURE in 50m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e97ac52327c4ba3b0f3cde2df6a6597 : FAILURE in 1h 47m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00e187875a5c48a9b8e5d67c38d662d6 : SUCCESS in 49m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/56406065e4674af99eeee29f4dd3f06d : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0ea740008494ec8b4c64db59f733df7 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9bfeb4fdd7cd4384b54b8beeb441d706 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b94d7cb7c65d4c23b2060139319332d8 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"96ec2dbe500ccb5cbed659417ccc5eddc8d751c5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-27 22:03:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"74b9b80ba751b3aad75b7e577186987c328ccf92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 01:48:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eee8798f1934906a182b11c22508671\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec7d6a9f8885442aa27e837f961521bc : FAILURE in 52m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a105a8e210b145f9a80a837f7c03b1a5 : FAILURE in 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"257c0e2281cb135b11a5b7ad7761cdc1565a7639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 02:45:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cd83f97f6324b448fa61946399fb111\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb2e1b6e4b143bc93c8b74cf6509b06 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d932bdbc284501afe60c8aab1e92ca : SUCCESS in 9m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf4d767ab2c845d5ad8da1ba9c9497c4 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e68499a72f43bfbf6c5f841776eed5 : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/591da9eea3de451fb6cfaadef95e6d63 : SUCCESS in 1h 24m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f44fc9a4573c40d2848fad0ee4616be4 : SUCCESS in 40m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5311bc62a38f4061b7bce6bfa34284c6 : SUCCESS in 1h 55m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bab89e577974525ac85747e3b761c00 : SUCCESS in 1h 19m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fcf0368ac6047e1b126f36ce27d63b4 : SUCCESS in 18m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3115e01c3c64bf2bf45f58e7ab3f1c0 : SUCCESS in 20m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c001b078c96a4fa283ffb057e53f7bed : SUCCESS in 7m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a1131aadc9e466bbb4fe057901fc38d : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/80408e01edd847208927dc5f3f625230 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3dc46f9d3a04f82a37d30cfdc250b6d : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/622da775d4d14401bec257ea05ea335d : FAILURE in 46m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe537c3bd91f49c9b54d764ef90b935b : SUCCESS in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8dec902cd0a44948bddf8be45788c272 : SUCCESS in 1h 51m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd2d98c7da614991ad409f40266584ca : FAILURE in 2h 34m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2be48abc41a64a02b7270fcca2fb1cb8 : SUCCESS in 2h 00m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bf984dea01f945fea1e36038831f716e : SUCCESS in 1h 44m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f5834bd3ed174a01973114cd46df6c9e : SUCCESS in 36m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2037f04d71c54c6c97bb4f4e808a0997 : SUCCESS in 1h 14m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/122397fbbb814f6ca4c2d1e11a90ffba : SUCCESS in 1h 38m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de24e35e96d145f186283ade8e54448b : SUCCESS in 38m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddf979d8516a4befb0def75c009232a4 : SUCCESS in 1h 18m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/08911c1f9fff4128b317dcc90aa0af50 : SUCCESS in 1h 23m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f41d7fc6b6644c3bb23db5af62e9fbf4 : SUCCESS in 1h 44m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c03d569a35054822a644068c7b688754 : SUCCESS in 1h 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/487f7ef11d6642b6afd85e5c60807245 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e85bd3e731cd386839a63b3529efdbf863949c9b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-28 02:46:46.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"5c40c4b97546c3bdd4991e1c42589d1b307c8bcc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-28 02:46:52.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"092906b5f385c3700a79352b30b976ada2d4f533","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-28 02:47:37.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"f576f4cb2b37fabcb15fe3e86598c1f4550d77ee","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-28 02:47:45.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"8647578c1ed173b379f2f5459afb6ad3acaf27a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-28 02:51:14.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_02_2025_02_45_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"9b0cbb86c6cd57fd658a84670e3f38b84cd31f26","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-28 05:40:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51016dcc1a95fb36b1cc8113cb407a208f2fa226","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-02-28 09:55:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7c422997a3c0699dc9f133bc9dace1a12bfc50e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 09:56:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3ae4fdfc8db457ddc6b731532abd4d3c845f866","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-28 09:59:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b32ba3c7bb2cc5e2d28f5f9291441443ad762ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 13:02:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd673945fcec4a65a5b4ddc39df9b0ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66eb0c718ed4aadbe30d57828116bdf : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2eee6c1e6e44b68bb92f5d030d4b9f3 : SUCCESS in 12m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ad8ca95e27b473683952471e9249dfb : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6c28c16fbb4bedbd243610568724a1 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a2896d1f37f14902a477802e49e2917c : SUCCESS in 54m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51a966b8e5b341c4907bf23be64b26f7 : SUCCESS in 1h 10m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da37ab585ce84d9fa0e7b3b313879ecc : FAILURE in 1h 35m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14285719f388438d85ac3621e53fc6bd : SUCCESS in 1h 18m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f83b9c3a8e64042b100cc69c43e6651 : SUCCESS in 18m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63fec41ec1bd47cba909ddbdc65c23f6 : SUCCESS in 1h 08m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0154ce63398844a2be0df71b40cfdb55 : SUCCESS in 2h 04m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/350351c287aa445598a9ed348ea6986c : SUCCESS in 1h 28m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f069202bd8a74ebcae136a5e2641dab5 : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":3},{"id":"6130666ac63a175ea5776ae17023fae802e1b724","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-28 14:34:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b6da13a9c6ae166e892f30e5eccfdcd9f96b101","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:12:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/775370d5d47e40e79ae81c5f79705b62\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6f486b2dc4c4b8b943a02253c0c9ca2 : FAILURE in 29m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9006a25a88c04a598e8f9e24706b4751 : FAILURE in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2711bae7cb55998e268f75636c79de6fe3655ee1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-28 17:37:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee11a6b95cbca6b642810f8a381e5839aa1faf29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:45:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/885036ea0780493c83461ac56183e82b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7d8b2e12b94d1881d4b7fb1a46b271 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df50f121702147bd8fe46029a0d61ccf : SUCCESS in 13m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94f9731e1bd04e2589d016034578d525 : SUCCESS in 15m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d90350452245a2af5b12ef3b19ef1e : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ed43336a0c6e486bb415dc576eb3788e : SUCCESS in 59m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc1e69ee29f64bbcbdaeaa20352e1aa8 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3858ce5acba84adeb1aeb9537b581a96 : SUCCESS in 1h 45m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9df0a600fceb46159403390f90a00e9b : SUCCESS in 56m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/291bf48728d64e28b10899aecfa698df : SUCCESS in 18m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7899a0ab3ee143c6a10046a9e7f4dd08 : SUCCESS in 30m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/885f94b3aa3f4405b439c9bcaa95d0cf : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebfc0679bfa54b35888486a7e095dd43 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ef10cb8185747db86ed42335fce1727 : SUCCESS in 8m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/db0ee87ede404dc88b251d093244104f : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12642514698a40f1a254bbc1f184adec : FAILURE in 43m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08c4ecc7e2014d81bb8e5ad5e1e2fb0e : SUCCESS in 8m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5316848d7fb04337bbdd598814a6be39 : FAILURE in 2h 08m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df3e2356e8a243bfa60d9ae859167a97 : FAILURE in 3h 07m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb231259678c4f3ca6be05f417b2775f : SUCCESS in 2h 06m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/54ef28a6be2341139833ef7d24837695 : SUCCESS in 2h 27m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ea0fa29454864bffafd2115cb3428e8c : SUCCESS in 1h 07m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c69696fe540746a18a4d5e54d2ee77ff : SUCCESS in 1h 00m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/833828cae2d04ff5b83ffbbc8f13a14b : SUCCESS in 1h 33m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1564eb02c9554d95947f1f1c8d857a58 : SUCCESS in 47m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/207b7d5fcd3f4e3597bcd5b9baaa6f10 : SUCCESS in 1h 34m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b6b9c91620104c9b8fea8d78c4a31c92 : SUCCESS in 1h 44m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57a71d361a734cfbbd51b8501722b64b : SUCCESS in 2h 03m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ff54d52580b45a59b1ad86ef8b5e86a : SUCCESS in 1h 28m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43570f008cd04b649fb5de942d7ee4e6 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3fd23ece0a5ba50e70e681cf900ac20abaeb2435","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-02-28 17:47:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ae7dfac23e4faf5e82cb3718307cbfaf0b612f0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-28 17:47:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ec6af547511c04ec559416c9237022dcc827b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:39:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a37c78374d349a5ae9350b8a1f2c5de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd95e06e487343ce9f8bc3800f58efdc : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd3036a7a50d48d4bf9ddad05f388ecb : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17e23cad165809c354bc52f01bcd2d19c8bb768e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-28 20:38:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5adc7d2e839a4f98b4dc1d28e1d52b89 : SUCCESS in 1h 23m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/26648fd61af241b7b45f1516d5dea16c : FAILURE in 1h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2402c8983468c5b02e77e0dec50b2d5c0c293370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 20:40:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b2b249164f4607ad0e560ae95250f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01d754d9456432681b622255de1ce81 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/267f11f945bb435799efd809760a2a3f : SUCCESS in 13m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2613ed3ebc3d4b3fa610b467101aec7b : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31ab09eb3bf4d1a9889f958f8182170 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/7ed1795d396347e4807919bb59e99342 : SUCCESS in 45m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/271cf3133f044474829103aacd121e0d : SUCCESS in 56m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55b8df3cb01b4fd58f8390abbf6ee464 : SUCCESS in 1h 39m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5874077dfc34905a888ced40c01bb2a : SUCCESS in 1h 03m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5763228e0f054f77bd65cbf8d71f5e43 : SUCCESS in 17m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ebc9f2f4f4e4b0a937087a403b70d5c : SUCCESS in 22m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2784aeea21b64acdb234b8a14bf9a7c2 : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b971c919f934e07899151d17a031499 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/98cb5f248bf84ed08d5eab731787bf14 : SUCCESS in 18m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7f6c93b419c4445f9e9857f594ef8b13 : SUCCESS in 13m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05f896da38e14583836e7b8b5bc12169 : FAILURE in 53m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ce25a77cfc343108e2ad94fb652b4a1 : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40894ac5ff154e06999ea85d5a45081b : SUCCESS in 1h 43m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/469efc9a94ce44938a7fcaa3ef57f985 : FAILURE in 2h 32m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f410abbb84c54828b210526d5780e377 : SUCCESS in 1h 54m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cde225d965c84ce08a3621d24ae981a3 : SUCCESS in 1h 57m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4fa439f984cb476b930b92223ae9d6d5 : SUCCESS in 58m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5df55ceed5b4178a8dd12c4c3f206ec : SUCCESS in 1h 13m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b0f7ae9d495443487faa3bb20cdf3b0 : SUCCESS in 1h 26m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c939cd724544423791750f22f3566cd2 : SUCCESS in 49m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/823bbe5d00d34a6292bde5b5fae0cc93 : SUCCESS in 1h 46m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/94e1f0d1ae924fe992b8d302ca3241c3 : SUCCESS in 1h 49m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/068c4644f2ed4eb2821cae56561fdc92 : SUCCESS in 1h 42m 31s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1801c441dfeb411ba11488b179564719 : SUCCESS in 1h 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7d6780e91c7474ea8d117f7017fa221 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e0ccaf83e387fb290670272c3002d707c655172e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 20:41:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e1b63344ab75673c1c3e423d2c6807b7a6124f07","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-28 20:41:41.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"eb7ef7275df736f4899b9fd8469a7ca1c2633d54","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-28 20:42:19.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"2c6c063bd693ef73fde84656ea982b3698f5f469","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-28 20:42:28.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"2b7d825915fb459b9a0ca94ddc8ae35cd90bfb3a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-28 20:43:05.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"1455c2f147e027641752b8610390d7e22a6412d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-28 20:50:08.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_02_2025_20_44_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"46e6999779078c5c5eee70898b2c7bd7901d781f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-28 20:54:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/d557a7c992b1442abe9f8b2fa7cab6eb : RETRY_LIMIT in 5m 18s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/69dfc22ce26440929d285470851aadf9 : RETRY_LIMIT in 5m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d6784ff72d79d6427663eba168a910b7fe29c7c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 00:12:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55ab8d822666478f84f5ce475c8b029e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1207d73b8d754cf2955db25dbc9af80a : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e02bc4c1b194107a50ea270d4fb1e33 : SUCCESS in 13m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a6682bde5244b9bbb851a6fde1c1ae4 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e060cb6814d941b9b66ef80b8ae0d7e9 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/60b12688f7544ee6a3a2bbe17bc45ce0 : SUCCESS in 1h 08m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5942a0f640de4e9e945fa7b2274a7148 : SUCCESS in 59m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea89cef4b8d64ef28afeb6a25508118f : SUCCESS in 1h 40m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a30b77b4521459aa6c42290e0ef1626 : SUCCESS in 1h 08m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/012bab0a7e964febaaa1bf62391a56fe : SUCCESS in 19m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e55af4a26f494b2e828a4c618713f7f1 : SUCCESS in 1h 17m 48s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/73b74ba9689445d8b9016cea6952106d : TIMED_OUT in 3h 03m 33s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810529a48c4c4f8794f30abbbc6b8f9b : FAILURE in 1h 24m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45548aabcbd842cebe0d0359b456127e : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4fc707d03a9b712d1719b5dcbd98527bc317e14f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-01 06:29:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"324edb48690170ca35fc43196f16652c8fd6f1ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 06:57:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41cc831251724539953a601a9f13927b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7735f95424c24ac48f3b50f3fa1b7307 : FAILURE in 25m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d0fd2780aa24c97a5b123225b7437f2 : FAILURE in 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9029ffabdf2b79c7f2310444b9ebb43d286db57e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-01 07:49:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fd07e88d095247eeb187b6e800367f4d : SUCCESS in 1h 10m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2e0b364f59de4018a48f06117f18033a : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7ab30b632f87222fed6360917aeb3d9bf9591a0f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-01 07:54:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b43a713a41c54fc59ac3a3b18a644a76\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/54bcf932ba0548d4a8bd0bc73725cae2 : SUCCESS in 52m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3cfdfbdf0a640f59e38cd629c0f6d58 : SUCCESS in 38m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c26b9e93e6884143a51725769152357c : FAILURE in 1h 24m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/257ae07c81254ee98cd317b32219d616 : SUCCESS in 45m 36s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39a7531af6ee4e47aecd8170bccfb242 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5df8d502c61d4534b9cb652d070104dc : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a41b90361387484dba0ee7b455346c94 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/48aafefe455947628155042e9a9506ec : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"4c566130bba320499e2befbb249198c1806fdd46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 09:08:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219fb305d6bc478b9ab9522e4713c933\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9b22f2186141cfab27244588220a5a : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b84a671af4749428073c65aad3f1343 : SUCCESS in 9m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82c6d7c3fb954a9cbf6038133a347d19 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa38f4cdbf7944fb93d49c034fc6e030 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/58a0d3d726034fa3952833b02c8d955a : SUCCESS in 59m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0eb9a2007e847c6a6dafd0338a00e15 : SUCCESS in 42m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e03fb1e638e4e6d9bb121436184c550 : SUCCESS in 1h 33m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cf7b3274a874e13952f4212063e4f12 : SUCCESS in 1h 03m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030268248bc2404c9b1ed106fee470fa : SUCCESS in 16m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/19265f2d8bff4866bb6d3128c30f1dd5 : SUCCESS in 24m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d6515ee7a1f845fc9ae7e80e34dfb2aa : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6fc60e8b963b4ec59c840bbceeed8968 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/259db6d848dd4f5889a82a3162d0b228 : SUCCESS in 6m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5de87cf53d6a4261ae1395db1316d6db : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a6795dc72df54adfb01b037789ddcc87 : FAILURE in 40m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/853d538570c84dfc8fe95cc21f6e6c0f : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5387c7a1d3f34b4e9cb827407e4e11f5 : SUCCESS in 2h 00m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9f035ba08e94534ad3181d12d453718 : FAILURE in 2h 32m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e89a691a6da34380ace72c3749115ab9 : SUCCESS in 2h 03m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9a267f83b0d8488ea6096e8a08881cc3 : SUCCESS in 2h 05m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6052a33b88dc417899b4f4d710a55268 : SUCCESS in 44m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f2768a7667d4be98ee416f9380bb44e : SUCCESS in 1h 18m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2e0f48354e74b4181dd50e87f84451e : SUCCESS in 1h 50m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d579e69f42b45b49bc944f1c235bc26 : SUCCESS in 56m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5588b5b87fa4b8bb5def739df2b2414 : SUCCESS in 1h 30m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5117094a421a4cb29f0a86ca44ae6d7a : SUCCESS in 1h 32m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c606b1ef52954506b374fcd33dd4ad3e : SUCCESS in 1h 57m 07s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c10b6c65beeb481dae2b98b3a6a4b0f7 : SUCCESS in 1h 14m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83ce53674486427c8a8883ee4ba43c78 : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ced8d5bd085def27976375768d61797aa23d12f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 09:09:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b0f79507a2c7b1e78310042425df8f7b12b6cca","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-03-01 09:09:41.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"ade75eb04d4fe43b810c7831459a2ec4e4048bc8","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-03-01 09:09:56.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"ee67a2be9ceae453c9576f487ad92250f032d2ac","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-03-01 09:10:26.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"dfda3e1c08ce6d83c01455e5573981ea53e00bf7","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-03-01 09:10:42.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"318817fb0e35950d8e925a4acf097ea8d136d237","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-03-01 09:13:57.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/01_03_2025_09_08_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"4f1d913d214f84e7ebea5c0b3c0a3d3a93a3cafc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-01 09:20:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/9ba32becdf264bf7981c7ce4f9271351 : RETRY_LIMIT in 5m 41s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/784acee20b8c49d9ab985c9fab018be0 : RETRY_LIMIT in 11m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"577e6107d5f541b48e452181266489e137b9ddb2","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-01 10:50:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c865bfd0401f38e6059b841de4f05836304382ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 10:56:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77e6b6d047944e3b820fd3d81c8374cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc340c25869a498da29bad3f37082867 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46c68bbb06354f7a8e02c1f851eea3d0 : SUCCESS in 14m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd152331c4e24963aa56ec5949ed63e8 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534a299db3ef4742848c84769d89224c : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8c1778281ac84e20ac646518d4d00ab7 : SUCCESS in 1h 13m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b921782e87864f3eb522fad38979c746 : SUCCESS in 51m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09ad40e25f1d402da4e53740a571a61a : SUCCESS in 1h 42m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c4e0e506084911a8db6d2ad1f8d75c : SUCCESS in 1h 08m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aebee60c063b400896d234012d89dfc4 : SUCCESS in 17m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8388ed9088fc45519f53be864fedf24a : SUCCESS in 1h 21m 28s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf3b803b2d834e678915b3c8094c67c2 : FAILURE in 1h 38m 00s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bdf23cb1bc94f26b7053725e2b1e4ef : SUCCESS in 1h 37m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73751718e35f4c25b23eda321ce24841 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":3},{"id":"47912abaa917e6c62a21dc1e636aa8abfc21f16b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 11:18:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ce1e68633942729231ff84f76dc93b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9af654c74e8e40a7837b6ba176ceba93 : FAILURE in 25m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42278a79d1db42e3b8f6f7209c33c833 : SUCCESS in 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02fd566ad5d4e0ad426219ff05f2dcf4439dd2e7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-01 12:08:28.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/3/20250301194537/NEC-V-ISCSI : FAILURE in 47m 48s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/88/931888/3/20250301194537/NEC-V-FC : FAILURE in 1h 22m 48s","accounts_in_message":[],"_revision_number":3},{"id":"69a4838aae6d3792f197bfb9567daa87bd66018b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-01 12:29:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cdc0a47e61c4482691c8547b7321207d : SUCCESS in 1h 16m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ec40a236770542e794b2831aec2c1582 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":3},{"id":"accc2f4b09887389d4b46dadac597dc86df923bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 13:05:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba3745e3a2b145b282f95536b3cd937f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73903a56f1f04c7aa203f9b937175187 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04293eb48e9b40ec88eb01169e5cfe5a : SUCCESS in 12m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edbae6629d4f489cb7493d0e6f00ed8b : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9945c91dd0466796b8f37ea6a40f2a : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/cf8cae4bc782412aa941124405701234 : SUCCESS in 41m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a2fe96048e74a70a793c88f7f5d1b35 : SUCCESS in 53m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63359eb064b041b08fd30f05d024e9d9 : SUCCESS in 1h 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/308ae242ebad4b24a0af089016b243ef : SUCCESS in 1h 04m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f487d258f05f450181e30ae9672a7212 : SUCCESS in 12m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/79f5ac6fd2354e038f77fba40d7d228c : SUCCESS in 24m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da7eb15a8ab241478b02c3cef6ad24d6 : SUCCESS in 7m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa000924dab744b4b36d2795988b53d4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/491c21f5bfb14ba79f9557127795a98d : SUCCESS in 5m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c0c2c0845821430ea8cc08a476800ca6 : SUCCESS in 8m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9160b498fb742368dd697d3d5b3a556 : FAILURE in 28m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e28d2e64bb24608ba192da4bcc93202 : SUCCESS in 4m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d24afd3bf194bca9a2df74a4caa5335 : FAILURE in 1h 44m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0a3e29182a3741f59eadcfc11dc25d79 : FAILURE in 2h 11m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e8f8aa9b5c594044bcab3312c09796c7 : SUCCESS in 1h 58m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/532c6e147c32492fabaec19b978700c7 : SUCCESS in 2h 02m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e7e717020fc84327b69d91fb87d42eaa : SUCCESS in 34m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3bc73de29b03408d833d6489109d9974 : SUCCESS in 52m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f718a3362a754491ba269f1e329a84b1 : SUCCESS in 1h 37m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/410c75dc3ab641b2bfc2ee8f5e963649 : SUCCESS in 40m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/01f693b5534a41db9bdd24bcbe174deb : SUCCESS in 1h 12m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0aacb56988048a087ae21dea45d5937 : SUCCESS in 1h 16m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2c7ef7847ad4f9ba27ab89038437731 : SUCCESS in 1h 38m 42s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98ddc2d719bf4e748334c9b548c9efe6 : SUCCESS in 1h 35m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b27226191443908c2f082490afa17c : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":3},{"id":"6133854f0fd546d0b3603586dcd6678b9d8e3571","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-01 13:45:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40f6e6900e35d329b1f9099b2330891baf202838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 14:01:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df0ec03c4ae4c71afe710617457d104\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/505143ec4c8440a4b6939a257e397ff0 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/feb503f208c848e8956973e2b26d5550 : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69865dff431fd3a71b41d738bb8ff98ca769267d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-01 15:11:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf05d066dd0a4ccb9f70c14a00f5a28d : SUCCESS in 1h 24m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/367063a5126e416ea02fcfca922f462c : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9b6443edf3d19ac2508af558bf16ac1193989d84","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-01 15:39:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/39e61dd6eccd4ef5bbc9890b6e11f335\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00a2ca9532b74d6998286f8035e6b1ee : SUCCESS in 43m 11s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38aed91e82c544c399ccf2a829f19132 : FAILURE in 41m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/913c71b0b31a4c9a8ac78e412c1e13a7 : FAILURE in 1h 53m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e68d232830b245a2bb643a92a249a9e9 : SUCCESS in 38m 40s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c91eb5dc0a24d39adde2be817ceb90d : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37e9287bde2c4fd8aac1ad6b21f5479a : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fb1aed3aaf142b28eca47f70a32f9f3 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bbc6535f64e148fa96a00fef0b2be757 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"cc265b6812bc7ad226af5d4c8cebbf6a03a9c906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 15:50:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52a12c6235fe437d8813311790428295\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e32ef75cfe74754b42238225274b5fe : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc24dc02e5b540a39f990d08aa02f4f9 : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c34bca4f9384a9aaf8080544b132ff4 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ef835684df4a08bcde97d2bd4592f2 : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d324dafdc5fb42ff84e189658573c8e6 : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8290f917c7b4c7a8da3ec261744b340 : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8bbe2c4a1804df39fdee6342e34ed0d : SUCCESS in 1h 44m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0b7c60a7d1440d8a58c0cd733f25698 : SUCCESS in 1h 05m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eb5d83cc16a42e19bfec13a614ecd85 : SUCCESS in 16m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dfcf4e10689342b4966c5cf2fa7e0164 : SUCCESS in 20m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be68096d37f24316ae39f426bfa937b5 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc083299a365459c8784966f8c9d031f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9478f20f94734fe59d8ddf5f25b41fd0 : SUCCESS in 5m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a1df3f5f24fa44e2bed6de696e5be099 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2412ecf7140c4c599dc82dc4f02868f0 : FAILURE in 26m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b19185a49004b6b99228b4a70be25a6 : SUCCESS in 3m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/08f0f4f38b3440b0898108f0baf215f2 : SUCCESS in 1h 52m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df6760c38ba04ba9aacb1e379169d1f9 : SUCCESS in 1h 59m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db68fe34584d47538931cf7f3c3269e2 : SUCCESS in 1h 51m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d2f759dc3c7b420b99845ed6ad8dd5b5 : SUCCESS in 1h 35m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6fd5460072cd4284a6c0fb8a0a44e65d : SUCCESS in 24m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ef5db067e7149f682e75c7712262bb0 : SUCCESS in 55m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af7972876c8f4b77a51737ac30aa6f21 : SUCCESS in 1h 14m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/85e3637054e545208f2c3d7be0d2a247 : SUCCESS in 34m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbd7d8a131db4def844407d37f39cc7b : POST_FAILURE in 1h 48m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/49dbf5ff265e474e98721a83a2e620dc : SUCCESS in 1h 59m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e2f9b41b1cf4d68a631e9982ea0a222 : FAILURE in 1h 26m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93f112c7155d4ace81cabf7fa659e769 : SUCCESS in 1h 35m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d9b962bfcc8483ba8ce1a3687862d09 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f8f584e3e9a5f1107ce74ae27886a0b9a4330400","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-01 16:10:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ef664e0c9078e7f452b0521a6c6471d4a6f70b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 16:39:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bde22aece2d4d78a51e7235cc16f0fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e4efdebe5d34633a3c82056a7353381 : SUCCESS in 25m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a545d33bddb247e2972519d45823692d : SUCCESS in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"161e51cc438e42b9834cb6f0d343efb638404577","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-01 17:29:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/094ca6fda34047e9af42595a9c2e02aa : SUCCESS in 1h 16m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f87dd14870946f28ec4720f3045fd7e : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d93bfd3b75398070026c278c30343cd49d9959bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 19:06:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa01cafd1eb041f59e9fc4c7bfd1969c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed436c5fd5a1449c845c7750169c2a69 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fc32e0e9d534fb0ae347a6279bfe5cd : SUCCESS in 8m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1925032437e40fdbd705fb489d6eb60 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfcc588433e443080436f227ca2daf3 : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8a92a7b759304da1bbdf18adcdff4345 : SUCCESS in 48m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e74dcd8e10e3482cae68222e6bb3f3b4 : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ce6a032e83b455d8aad0523b64ce1ff : SUCCESS in 1h 16m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26becc92c43e4de9b3b4341ec32fb0ef : SUCCESS in 48m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe7f1d70efe4f6abcbd467b7a33b543 : SUCCESS in 17m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d60deaf14fe416e86e1d9460a885824 : SUCCESS in 28m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/73bc1e4d2a2a459984aac5d81a711ccd : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11b6d9daf2e54f03aedce18037e54bf3 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a74a124088664feaaddc83b2addd3386 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f8c272faea094703a754cf9a7fc62f57 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82f48d3b82534918b40aa3e138c622bd : FAILURE in 59m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97ae32f00cf74cd581f3475768d10e85 : SUCCESS in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c1397056d814f0197644079de86fd53 : SUCCESS in 2h 04m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d73915d20dfb41398f67555920a7dd5b : FAILURE in 2h 36m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4e70f95ffe647b3af2f408002a52e2a : SUCCESS in 2h 07m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a7729eb8a8984c5692e1966421a9e2b6 : SUCCESS in 2h 12m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/881320c505404fc6a8c91a4b03c56afd : SUCCESS in 25m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4192bc50fe8f4945a58533efc05b4070 : SUCCESS in 1h 17m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/34819366c99b4eee8de098e2ce26b514 : SUCCESS in 1h 34m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/804ba5e5fc2a4724b8254c6265231a36 : SUCCESS in 43m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/18e8c8c39403411b91a21a432f5bf813 : SUCCESS in 1h 39m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1abae17652144faea9981c2153395e91 : SUCCESS in 1h 59m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f56f11675ca4656bb63caeb9b09f459 : SUCCESS in 1h 50m 02s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/870c5cf55b3b42c3963cd818318cfb9c : SUCCESS in 1h 33m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5c39f168e35450681d71cbc8bb70745 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":3},{"id":"89be9fe817a59c15d8b29139ed95d7f90499fd61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 19:06:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b3ebc70f0cb0b70d29fd3df3090d500bd4a5b2a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-03-01 19:07:25.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"706634897aeff8c1c3237dffc62dfeba680fea55","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-03-01 19:07:43.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"08f7d642d78e7c7d2edf6ef208349a8a765e18d6","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-03-01 19:08:11.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"f9751a3517df8fe8b82d86b9f94ccbded1c052e9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-03-01 19:08:36.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"8622dac428b216823b5b8881f408ae6f74930cd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-03-01 19:11:33.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/01_03_2025_19_06_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"18658d2f4c92ca2f4afed7a9a4af012b0b7c5f05","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-01 19:13:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/a99b843dde77402da87e22baede70f08 : RETRY_LIMIT in 5m 39s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/6ff169337f464777a63f0b0cb739a6b7 : RETRY_LIMIT in 5m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"3f20690270c08b14a38b100cdac1a68ad6defa80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 21:08:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f76115cced1473a9981e9180a1e100a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502c7dfd036d432bb555f336465bc06a : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc54ccbbd01944e6aff9664d2e36be62 : SUCCESS in 12m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b8d27c1cf4e4d48b81cd85391e4702d : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790ed4eeb5e04c3a8d81a880a3fd3640 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5cc679ac7ca04ee1864a10e1be66679b : SUCCESS in 1h 02m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9bb295979c647329a79b77901886a01 : SUCCESS in 1h 02m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00617944c5544c2e9b63b531862f5481 : SUCCESS in 1h 32m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3adff832349446a1968652927e88479a : SUCCESS in 1h 05m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e664cf6ff05a49ae86e36f6fc9683780 : SUCCESS in 18m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d3b3ab714fd40d1a6cb9799c59f5626 : SUCCESS in 51m 13s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ce003684a774d89a29959e341bc5937 : SUCCESS in 1h 56m 21s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dffbf4d199684514a4cb6eb8f78fa115 : SUCCESS in 1h 20m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/879ce5673b1445429dafb8c9fe672670 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f0b42050591cd62842160e55d99aeea9e0063127","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 21:08:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7b71677db193186ce7bb65dfd6881acba1aa7f6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 21:09:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc07e2f7b27448695b96377062a3e69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7709d76e2a144f0968d02fd61a1c951 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fbed1c26e264af1bb5df2125165a02b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3},{"id":"14e5e61e44c4edec184aec5ef992061a8e4482cc","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-11 15:46:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/931888 : FAILURE in 33m 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c05d12875fb989fc786b3c91a1ea1ce722d0f382","revisions":{"b4fc9c5942bbff63fde9f458e9e5e24a2e4f503e":{"kind":"REWORK","_number":1,"created":"2024-10-09 08:46:40.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/88/931888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/931888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/931888/1"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-10-09 08:26:24.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-10-09 08:46:22.000000000","tz":0},"subject":"Dell PowerMax: Added exception handling following the masking view connections REST call.","message":"Dell PowerMax: Added exception handling following\nthe masking view connections REST call.\n\nAdded exception handling after the masking view\nREST call to retry fetching the host lunid if\nit is not returned by the REST call\n\nCloses-Bug: #2081742\nChange-Id: I9e8e5f38c8ab8804c5e7ae4ad5e0f650768e93c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4fc9c5942bbff63fde9f458e9e5e24a2e4f503e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4fc9c5942bbff63fde9f458e9e5e24a2e4f503e"}]},"branch":"refs/heads/master"},"e96f1caa88be9f19ee63060636d1b38f7872bd79":{"kind":"REWORK","_number":2,"created":"2024-10-10 05:15:16.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/88/931888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/931888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/931888/2"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-10-09 08:26:24.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-10-10 05:13:58.000000000","tz":0},"subject":"Dell PowerMax: Added exception handling after the masking view REST call.","message":"Dell PowerMax: Added exception handling after\nthe masking view REST call.\n\nAdded exception handling after the masking view\nREST call to retry fetching the host lunid if\nit is not returned by the REST call\n\nCloses-Bug: #2081742\nChange-Id: I9e8e5f38c8ab8804c5e7ae4ad5e0f650768e93c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96f1caa88be9f19ee63060636d1b38f7872bd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96f1caa88be9f19ee63060636d1b38f7872bd79"}]},"branch":"refs/heads/master"},"c05d12875fb989fc786b3c91a1ea1ce722d0f382":{"kind":"REWORK","_number":3,"created":"2025-02-27 19:14:19.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/88/931888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/931888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/931888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/931888/3"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-10-09 08:26:24.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2025-02-27 19:11:32.000000000","tz":0},"subject":"Dell PowerMax: Added exception handling after the masking view REST call.","message":"Dell PowerMax: Added exception handling after\nthe masking view REST call.\n\nAdded exception handling after the masking view\nREST call to retry fetching the host lunid if\nit is not returned by the REST call\n\nUpdated the retry count based on trial and error.\nThe maximum observed retries reached the 8th attempt,\nso we have set it to 8 retries.\n\nCloses-Bug: #2081742\nChange-Id: I9e8e5f38c8ab8804c5e7ae4ad5e0f650768e93c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c05d12875fb989fc786b3c91a1ea1ce722d0f382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c05d12875fb989fc786b3c91a1ea1ce722d0f382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
