)]}'
{"id":"openstack%2Fcinder~931976","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I009d0748fdc829ca0b4f99bc9b70dadd19717d04","project":"openstack/cinder","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2024-10-24 16:17:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I009d0748fdc829ca0b4f99bc9b70dadd19717d04","subject":"RBD: Flattening of child volumes during deletion","status":"MERGED","created":"2024-10-09 17:01:36.000000000","updated":"2024-10-24 16:18:55.000000000","submitted":"2024-10-24 16:17:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931976","cherry_pick_of_change":835384,"cherry_pick_of_patch_set":44,"meta_rev_id":"d143544607294b01957861da38ba700d638eb193","_number":931976,"virtual_id_number":931976,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-24 16:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2024-10-18 20:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-24 14:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-24 14:32:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 17:01:36.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-10-09 17:02:00.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-09 17:19:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-09 17:29:03.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-10-09 17:29:04.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 17:30:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-10-09 19:08:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 19:14:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-10-09 19:27:53.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-18 20:31:43.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-10-24 14:32:23.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"43664ede47962d577d7b6b2e7d790f380e9559b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-10-09 17:01:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6cf3245df588b608b57082b7cba22fece83c2eb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 17:02:00.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/721/","accounts_in_message":[],"_revision_number":1},{"id":"47793e5b8e127cec97e969901d10890d3620203b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-09 17:19:39.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {Ma... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"868b7f84e924cf251c20bc1e718d8ae04b500cd5","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-09 17:29:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/931976 : FAILURE in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a99e66487e46fd2b08a17d163868c1e36e71ff10","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-09 17:29:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/931976 : FAILURE in 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1df9b71c05017f54a059ac4bf705921825b434fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-09 17:30:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/931976 : FAILURE in 27m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f44983874ac4124f67dfa09e73df54adcac51d8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-09 17:36:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931976/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/931976/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"28dcd5ceac49e046cccfddc672d49cd66387689e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 19:08:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5411461ef239442ea51ae6a2b3db58bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f56fd458654402b8211892cbe7cd16e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575d9c97ad8940baa0d2194e49f818de : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d253be5ef24adeb19ef04b647c30b0 : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1af4e54c6394b4b9741dc708870f42f : SUCCESS in 9m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15e0c54db19c4187810f6a66ac30fb46 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d74ba1117d5435c8a63d4600a454ffb : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a44db616e58545779f3536bd02e258a1 : SUCCESS in 55m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49c979b3217e497790572874438f4319 : SUCCESS in 1h 18m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42e0855647fb46819282d477b73f0560 : SUCCESS in 40m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f010778eef97446893872be143cfb7a8 : SUCCESS in 18m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e44074db41a4b87963e58ffb637d109 : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5f20ee90c9c4dde8619b3db15d539e7 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4a5d8d84bcd4a208fc2350978611546 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e78a60fff37f41f6985868d22505e37d : SUCCESS in 5m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5eee344c49d74de59208e260d3c2193a : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3088fd8400994f36a14a1d45ad60fae3 : SUCCESS in 31m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/46f5a7db1b754e48a8a1bae67c433a49 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/087ec12819f8407d86c244f6aa81d5d6 : SUCCESS in 1h 52m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de33e758370e4faaae97e80c654c8abc : FAILURE in 1h 53m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7487ead223f94d41bf009f9b495a99c2 : SUCCESS in 1h 38m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/530aaf8e0525455cbaefd3d2ac16a18e : SUCCESS in 1h 52m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a91c3d5207b4df7ab15507e548575d7 : SUCCESS in 56m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e97e9ec37274a6d9998416c6a5fe12f : FAILURE in 1h 42m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c3d7fa4412914d61bb174a3717957b0c : SUCCESS in 41m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16453915dc1e4f51a4eabe87448cb742 : SUCCESS in 1h 11m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fb39e719a1c543afb47ac6d249f721d8 : SUCCESS in 1h 25m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbd68d74d0e54461b035a5e87315441a : SUCCESS in 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ceba49951ec4f358091a31a132377af : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":1},{"id":"af115e2afb92b83448762f56002e94814dd40b3f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-09 19:14:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_10_2024_19_08_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"074f32b6087e5d0b9e4a7f26438103087d2620ff","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-09 19:27:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/931976/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/5757acc/ : RETRY_LIMIT in 8m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/931976/1/upstream-check/cinder-tempest-plugin-ontap-nfs/1a1e2d9/ : RETRY_LIMIT in 8m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2daedebaa3c55bd6de5594d804c76c56032377f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 07:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e55a567a8524806a322115e58a73e4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87e39321311e4ca39fb7a1bc17846c22 : FAILURE in 21m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cc1e70cce9442e099f124c4eb3f201a : FAILURE in 20m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1f1269563a64f29aaf06e9bab7bc497 : FAILURE in 20m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba27bcf1fc724f7fa696efdf3a5ae1a3 : FAILURE in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"561ef79328db52c72f9bbe053d6780b5e49085ac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-10 07:46:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/931976/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/657388a/ : RETRY_LIMIT in 5m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/931976/1/upstream-check/cinder-tempest-plugin-ontap-nfs/3dadc69/ : RETRY_LIMIT in 5m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"efdc52abcafa96eee0faf4f65bbe9a5a00da8d3a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-10-18 20:31:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66853763b71f1509a04b237ca9425ae2e49bd6e1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-10-24 14:32:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19cdfcafe190e7ecddc7db876db6609724c7aba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc3854be980f7e0efe4d7d4dfe269dc011e7a029","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-10-24 14:33:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d4d3b70c5d6c5a04fde7e66c8f70c960cbec68f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:17:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61eafb84631f43d696ee8f8792e72439\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7419de6b6fbe44548c478e9254f6b8ba : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a12570fa3348348d747968e23dd9eb : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ae8919964504dd4a574d50b8241c4ef : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75761073b940489c8679a2a6edbd8882 : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5916236127444a8384f685ddfe9c72b6 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/790eb95c0d40429193688656e58c40ff : SUCCESS in 50m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/391c62bf52e8474db2de2b9fed2fb705 : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3d80ada72fe4494a780e036dd1d3a98 : SUCCESS in 55m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d066f2e90bd24a578c566a213ca000ea : SUCCESS in 22m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1683553686194d909f77adc0e3473be3 : SUCCESS in 48m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e0d370da58142cb8d46e3d378c3fa08 : SUCCESS in 1h 42m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b92e9018e272452ebea3664c1e4c1607 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e99ce7336ff9ac440f1d1f168f5c1a407f3d891c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:17:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d143544607294b01957861da38ba700d638eb193","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:18:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb33ddca98e14f35a7fcb741ad98df62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b85a95e898149dfa5a7067f70992912 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63d558e31f5b43d992bd57515efae208 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab92378c47b4add7744b8841e096d0d99830b0f1","revisions":{"ab92378c47b4add7744b8841e096d0d99830b0f1":{"kind":"REWORK","_number":1,"created":"2024-10-09 17:01:36.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/76/931976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/931976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/931976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/931976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/931976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/931976/1"}}},"commit":{"parents":[{"commit":"ec37accf6206a5054a9219f51d74128a9c5d3b4e","subject":"Make default-types APIs compatible with V3.67","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec37accf6206a5054a9219f51d74128a9c5d3b4e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-04-20 16:08:31.000000000","tz":-240},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri@canonical.com","date":"2024-10-09 17:01:36.000000000","tz":0},"subject":"RBD: Flattening of child volumes during deletion","message":"RBD: Flattening of child volumes during deletion\n\nThis patch allows delete_volume and delete_snapshot calls\nto fail less often when using RBD volume clones and snapshots.\n\nRBD clone v2 support allows remove() to pass in situations\nwhere it would previously fail, but it still fails with an\nImageBusy error in some situations.  For example:\n\n volume1\n   -\u003e snapshot s1 of volume 1\n     -\u003e volume2 cloned from snapshot 1\nDeleting snapshot s1 would fail with ImageBusy.\n\nThis is fixed by using RBD flatten operations to break\ndependencies between volumes/snapshots and other RBD volumes\nor snapshots.\n\nDelete now works as follows:\n  1. Attempt RBD remove()\n     This is the \"fast path\" for removing a simple volume\n     that involves no extra overhead.\n  2. If busy and the volume has child dependencies,\n     flatten those dependencies with RBD flatten()\n  3. Attempt RBD remove() again\n     This will succeed in more cases than (1) would have.\n  4. If remove() failed, use trash_move() instead to move\n     the image to the trash instead.\n     This allows Cinder deletion of a volume (volume1) to proceed\n     in the scenario where volume2 was cloned from snapshot s1 of\n     volume1, and snapshot s1 has been trashed and not fully\n     deleted from the RBD backend.  (Snapshots in the trash\n     namespace are no longer visible but are still in the\n     dependency chain.)\n\nThis allows Cinder deletions to succeed in most scenarios where\nthey would previously fail.\n\nIn cases where a .clone_snap snapshot is present, we still do a\nrename to .deleted instead of deleting/trashing the volume.  This\nshould be worked on further in a follow-up as it is likely not\nnecessary most of the time.\n\nA new configuration option, rbd_concurrent_flatten_operations, was\nintroduced to limit how many flatten calls can be made at the same time.\nThis is to prevent overloading the backend. The default value is 3.\n\nCo-Authored-By: Eric Harney \u003ceharney@redhat.com\u003e\nCo-Authored-By: Sofia Enriquez \u003clsofia.enriquez@gmail.com\u003e\n\nCloses-Bug: #1969643\nChange-Id: I009d0748fdc829ca0b4f99bc9b70dadd19717d04\n(cherry picked from commit 1a675c9aa178c6d9c6ed10fd98f086c46d350d3f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab92378c47b4add7744b8841e096d0d99830b0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab92378c47b4add7744b8841e096d0d99830b0f1"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
