)]}'
{"id":"openstack%2Fcinder~932257","triplet_id":"openstack%2Fcinder~master~I50ac742749b3a58cfd143ce92da72118e497f2af","project":"openstack/cinder","branch":"master","topic":"tox-envdir","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-15 15:56:41.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-10-15 13:24:53.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I50ac742749b3a58cfd143ce92da72118e497f2af","subject":"tox: Drop envdir","status":"MERGED","created":"2024-10-12 16:28:13.000000000","updated":"2024-10-15 15:58:07.000000000","submitted":"2024-10-15 15:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932257-tox-envdir","meta_rev_id":"af308e8a239f41aeec2b94760f019ea419456448","_number":932257,"virtual_id_number":932257,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-15 15:56:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-10-15 12:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-10-15 13:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-10-15 13:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-12 16:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-12 18:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-12 18:21:36.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-10-12 19:21:04.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-12 21:44:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-13 02:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-10-13 04:54:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-10-14 00:52:09.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-14 13:38:12.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-10-14 16:49:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-15 08:31:57.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-10-15 12:48:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"810dfdb1b12228a8d2923a6a7d68db29ffcf2986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 16:28:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f08fc21284d2c4920177ea677ad7c6336484fca1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 16:51:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/062f471437a14c9c9b92e35837455048\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f622e7343c4ef9b21da972d6fd4a6d : FAILURE in 22m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8157a69b58c745bb932e5c56800be55f : SUCCESS in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df34d00a3205b1b2c194f09a58b2b8eb34bddb6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 18:15:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7614fd5ac3ea4d37a3884b7d55bd1d74\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/40c8324b07524a2f83de9b4d08855632 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06af2d3d257c42b4898f66ebf5a8ec74 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef42bf8c3654834b6a4fe1789fa461f : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae5c5b55de2546f6b03b68377a400451 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da386b9bc0b44494ad1aa4ae9d1b9ad4 : SUCCESS in 14m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68cd694f7e7c4c178fe193beefb397ce : SUCCESS in 41m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1db395415b94ff68e9f670dc372001f : SUCCESS in 13m 41s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5be3eb9d0ed9462ab6fbffb3e1e71d48 : SUCCESS in 4m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82acad5b72124a9fbbbfa9eaac476e9c : SUCCESS in 8m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/78158b27f6074367a70c6ee132c07f0f : SUCCESS in 8m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/90818b0738d44cc98c7d6f2840bc34d1 : FAILURE in 30m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11b8b86b154e48889c4ff939bce2d163 : SUCCESS in 7m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6afcfe21e8dd4422bbb9f27911b4ec62 : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f07b7aa2419683d96c3537ff225d618d85d9346b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-12 18:21:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_10_2024_18_15_55\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"68337c0878e4c450deeb0826c47b6d327f1012dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 19:21:04.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/731/","accounts_in_message":[],"_revision_number":1},{"id":"8aa704791e96069ed675c5ca45ec04809b0e3569","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 20:21:22.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/932257/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/932257/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"6986ecaaecb0a2ea1eb709f16a61fe37b66fc199","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-12 21:44:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d172ef6290474944b299dc0869a74f27 : SUCCESS in 1h 08m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f0eaf88a27e547219ca3190fc0af1758 : FAILURE in 1h 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5f70bcf1eeec0fe2ec4ff25794376bd9c9b44b59","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-10-13 02:00:09.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/57/932257/1/fc : SUCCESS in 0h 53m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/57/932257/1/iscsi : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5013f13a6659b523994c6feaf5488cb199313b1f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-13 04:54:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerStore_iSCSI/bffec9f/DellEMC_PowerStore_iSCSI/1815 : SUCCESS in 1h 25m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerStore_FC/7777421/DellEMC_PowerStore_FC/2013 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerStore_NFS/a54a840/DellEMC_PowerStore_NFS/1508 : SUCCESS in 1h 12m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerStore_NVMeTCP/22fa117/DellEMC_PowerStore_NVMeTCP/1367 : SUCCESS in 2h 18m 18s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerStore_NVMeFC/5e66506/DellEMC_PowerStore_NVMeFC/332 : SUCCESS in 1h 13m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"81eca22c9bb8e313011f36b579a0f2de2a310c0f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-14 00:52:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerFlex_3.6/547a8cf/DellEMC_PowerFlex_3.6/2233 : FAILURE in 2h 44m 17s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerFlex_4.x/b092307/DellEMC_PowerFlex_4.x/1115 : SUCCESS in 2h 02m 34s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/c2e5b1e/DellEMC_PowerFlex_4.x_NVMeTCP/1041 : FAILURE in 5h 54m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1940c7a7735aea85323c606e5ac1f4a61e6e609f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-14 16:49:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_Unity_FC/23fb6c8/DellEMC_Unity_FC/2734 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/57/932257/1/check/DellEMC_Unity_iSCSI/64ffebc/DellEMC_Unity_iSCSI/2731 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac7dffc4ec2e4e9fd436e10a8211a1cdbefdee70","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-15 08:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/932257/1/20241015133349/NEC-V-FC : SUCCESS in 3h 54m 30s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/932257/1/20241015133349/NEC-V-ISCSI : SUCCESS in 3h 58m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cebf959863be651c259b6e3582f64dbb4fabcf64","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-10-15 12:48:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61b5ff876881cc5d0c943d117e554e50a2493215","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-10-15 13:24:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"540b382b219d042bb64ffd94e9cb4741b61ccb4f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-10-15 13:24:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2cf618500980cd5f9691400f782a6af708ecf52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 13:25:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbb9ce3cb4d41b0b69108313e65301251f2ab941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:56:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16eb6772a20c46f7bb61abc83c47d1d9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/19fc1b95b2db4c6f80545cd0d479f935 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603c35cf625e4b12abd6809c8c0fea2e : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0c5ab0929f42d6be9fff500f9575ac : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c99ccb44b0fe451da3d185a123fbac1d : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc4c999ca09448d92d756de9bbdce63 : SUCCESS in 13m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1dda23d9184418d8170a7932db54b6a : SUCCESS in 1h 05m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ba6a7bb8db4a03adbb38f2126cb515 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3db581ab987f958a82e37f6164f8a9c690cd3ffe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:56:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af308e8a239f41aeec2b94760f019ea419456448","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:58:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d4d878a156467fb76c5e14fdaf4dab\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5e4b48711ae34bf99fb80312cf5b19f4 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb0b859c2194d3cbae505f310b931e6 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b146a3c823490a99d698cd5122b126 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c5b35a1a96b10fa3af657e5ca83004278096ddf","revisions":{"8c5b35a1a96b10fa3af657e5ca83004278096ddf":{"kind":"REWORK","_number":1,"created":"2024-10-12 16:28:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/932257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/932257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/932257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/932257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/932257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/932257/1"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 16:27:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 16:27:59.000000000","tz":540},"subject":"tox: Drop envdir","message":"tox: Drop envdir\n\ntox now always recreates an env although the env is shared using envdir\noptions.\n~~~\n$ tox -e genpolicy\ngenpolicy: recreate env because env type changed from\n{\u0027name\u0027: \u0027genconfig\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027} to\n{\u0027name\u0027: \u0027genpolicy\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027}\n~~~\n\nAccording to the maintainer of tox, this functionality is not intended\nto be supported.\nhttps://github.com/tox-dev/tox/issues/425#issuecomment-1011944293\n\nChange-Id: I50ac742749b3a58cfd143ce92da72118e497f2af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5b35a1a96b10fa3af657e5ca83004278096ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c5b35a1a96b10fa3af657e5ca83004278096ddf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
