)]}'
{"id":"openstack%2Fcinder~933216","triplet_id":"openstack%2Fcinder~master~I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8","project":"openstack/cinder","branch":"master","topic":"bug-2107469","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-11-04 10:42:55.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-05 09:42:55.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-08-04 12:01:37.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-30 07:50:36.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8","subject":"Add finished_at column for backup","status":"NEW","created":"2024-10-24 09:01:02.000000000","updated":"2026-08-04 14:45:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe8f18112999a5971b157af9ae0acea2c8af195a","_number":933216,"virtual_id_number":933216,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-04 14:28:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 09:01:11.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-24 09:55:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 09:56:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-10-24 11:05:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-24 11:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 12:13:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-25 16:30:02.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-10-25 20:07:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-26 00:38:37.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-26 15:36:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-16 13:00:53.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-16 13:34:34.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-16 13:36:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-22 18:51:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-23 18:34:35.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-24 08:38:33.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-24 16:51:19.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-25 00:58:54.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-04-25 09:10:54.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-05-05 09:42:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-05 09:42:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-05 09:43:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-08-29 11:20:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-10-24 06:06:25.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2025-10-24 09:49:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-25 03:46:03.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2025-12-17 00:06:54.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-18 21:50:26.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-02-18 23:31:29.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-19 08:53:25.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2026-02-20 16:20:29.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-20 16:35:39.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-22 00:39:08.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-06-13 00:06:16.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-13 00:57:06.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-14 13:38:13.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-04 08:36:03.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-04 08:38:09.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"}],"messages":[{"id":"9ac5a83ef8aa5bd38731b00fe7fa67dbbd8deed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 09:01:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1ad708f6122e8f714def24004219f1ef55d7d96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-24 09:01:11.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/804/","accounts_in_message":[],"_revision_number":1},{"id":"4db09dd3e28c6753209b0e9cc96240bde1bfe715","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 09:01:16.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"bc4a0e7ec21593213fcd12c4fb8199698b43ea13","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-24 09:07:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"f087915b65ce68e4ee77c964c705a992bafc6b2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 09:55:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe6057aa7d2496e979aaa3c541ddbcf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cf93189cbf04cc8b88a5c2a58291e77 : FAILURE in 52m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57205a7a3cc141fa90eaacea06ef4c50 : FAILURE in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b67d9308e46b4e91de8eb6e346fef1f837755775","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-24 09:56:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/749/.","accounts_in_message":[],"_revision_number":1},{"id":"81822939091fa148571aa081ee7e863853593a62","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-10-24 10:28:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1323/.","accounts_in_message":[],"_revision_number":1},{"id":"fc46b8d9bee6a53d4304b141bc9f659f3cfe150e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-24 11:05:51.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e9cefe399074466895531dd1343d0341\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8fecaab6ed04f46b1ff317afd656656 : FAILURE in 40m 20s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff1b3c3ba1c94585b8cc23e2c45bc600 : FAILURE in 1h 19m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ec7f19082fc4940a422d519f24ccf7f : FAILURE in 2h 04m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e37737641f548b1be75aac13f524e70 : SUCCESS in 40m 29s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da312a2bd3664a5eb679174fcbc2ec24 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69cd8094c3b842d1b182ada9b05a95c0 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/890a75f061684f3cb85eed4ca22ee7ce : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d2adfd48cbff4b3ba50bd7864ef144b2 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"ba648ecfa4f4f45aab26cd7daae0f502e2ad6f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 11:59:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6d2f8424d964728b22d3c449e24e63c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a92ebfc0f9634691b86d893968ddbf50 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429ab0cec5b246b9827efbfc08784168 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/598c98af4ab946c3b0c63fb888597d79 : FAILURE in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e6ec6106b994e0b9e9b5ccff91a107e : FAILURE in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a0ddc2d18b468e966f010bffc083c7 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/5d63966c51e34c4fa551a2f27d309e3d : SUCCESS in 1h 05m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f3d101732a344eca1bf6320c82b4eee : SUCCESS in 55m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c294d15f6ef94c5f8fcd1d82eeea3445 : FAILURE in 1h 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ea2f6268fb417382bf456c327570d7 : FAILURE in 45m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06aff85fb58f410da6aab5e33ee523a1 : SUCCESS in 17m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f3507f4690e34fe3a7736fad48b70b32 : FAILURE in 13m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c574d6d677c347fb917382a39a2054c9 : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e53bee194ae044af9f509904df3330d0 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a4d01ab2e384c2083c1a65cc2061f10 : FAILURE in 5m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/44af1deaaa974ea8b6b156f44c9d0ebe : FAILURE in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3275d97107764b5385e55e518873874a : FAILURE in 25m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a79dbc382b14734bd3b15d78b55ff61 : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44a80d0376f0490db26305301f23f66d : FAILURE in 1h 47m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c1afffd8d2a450181d7fb5d8265aba2 : FAILURE in 2h 38m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/49da4507c4cf44ca963602ba5cc4a830 : FAILURE in 1h 55m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2c2a4f66bf5845f8927c432e330d487d : FAILURE in 2h 53m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e4a6560025b545178644add56ad2380b : FAILURE in 26m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b0564ee5b6a424abd2206d5dc3d5073 : SUCCESS in 57m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/87e86f2d60f5412bb60b6ef6003dbf71 : FAILURE in 1h 51m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c386f569f5404a2c90449299196c2081 : FAILURE in 46m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4091b0bc75b5440587262f187fe1c1db : SUCCESS in 1h 40m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0dd6c95c94fc4a1ba008df2c79ff47d5 : SUCCESS in 1h 25m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/792a0b4767eb4b0ea1e9179a16123377 : FAILURE in 1h 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f66279c6d3d4a1fa314bc09c7bce10a : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e92dfdbc8d8522f01c057bd5cd59ebe3d139afd6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-24 12:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf14f5d34da0441eb544b217df75656e : SUCCESS in 1h 18m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/633fafb4e05b488cbca66c81690c36f2 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a09d94227ea4f91bf88e797f422a1d7ae9d7e0d9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-25 16:30:02.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/1/20241025153834/NEC-V-ISCSI : FAILURE in 7h 38m 46s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/1/20241025153834/NEC-V-FC : FAILURE in 9h 51m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7c96c738a60b0ff3f0b008d8d07b5a20e266452b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-25 20:07:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/933216/1/check/DellEMC_Unity_FC/43e49e7/DellEMC_Unity_FC/2790 : ABORTED in 3h 00m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/933216/1/check/DellEMC_Unity_iSCSI/d6fb7ec/DellEMC_Unity_iSCSI/2775 : ABORTED in 6h 00m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5afae652b349f485efafbd1bf020407876b4d061","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-26 00:38:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/16/933216/1/check/DellEMC_PowerFlex_3.6/cf9f210/DellEMC_PowerFlex_3.6/2294 : FAILURE in 2h 57m 00s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/16/933216/1/check/DellEMC_PowerFlex_4.x/a7be194/DellEMC_PowerFlex_4.x/1176 : FAILURE in 2h 52m 13s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/16/933216/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/c4c886a/DellEMC_PowerFlex_4.x_NVMeTCP/1090 : FAILURE in 5h 54m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"707e1635b02bc48846e09d62f33bcb8ae2c087b7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-26 15:36:26.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/1/20241026132147/NEC-ISCSI : FAILURE in 10h 08m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/1/20241026132147/NEC-FC : FAILURE in 11h 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b7608d09d78e7c529637599bf9111d88b7b7124c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-16 12:59:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14d2b3f0b455b2373b6ccf9f442c9fddb9968fdf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 13:00:00.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1718/","accounts_in_message":[],"_revision_number":2},{"id":"3022c960be70033210d23e4df2262c3e6186537d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 13:00:02.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"98e4151f4bb449c4528cdfbe8218c8e32a744a4f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-16 13:00:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"969fced4a51fc3e8b848644f37a3cb46d083acf2","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-16 13:00:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"eea2948f295e8e00a9351ec02b1f667a98ea84b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 13:20:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b71454236d08480d962259e0e128ede8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a709f40ce5c047db9cccd71db6a80418 : FAILURE in 12m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b54e9cc7ae4342688222fb0ffda0fff8 : FAILURE in 14m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55748e1da3bc4f98804edb6fe760dc3b : FAILURE in 18m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8255f2b824abfb5e5815bfe63c6761b0260d2c70","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-16 13:34:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 32m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e2d79000096cb2d5c81090dc067dc4203a751a24","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-16 13:36:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/933216 : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":2},{"id":"aa0bacc1a322deb05ea0d8a1bcc3a4bbb93d37df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-16 14:21:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/233795613c10491d947f35e80c903f2c : SUCCESS in 1h 19m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/869ec01086b0402db123aec6c63af36e : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"41bc86a6e3c5adcd305a233c6a144e65eb9543af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-16 14:46:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/47c63c8698ff4cdc9b5596be4780cb63\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08bab0596d9f4015a35fbd8258095df9 : SUCCESS in 39m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c9406517f9647cc9541378f921be816 : SUCCESS in 45m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6f88525abdca4b499a31cdcf82503a6b : FAILURE in 1h 45m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19472466c641466c901dcbce905e555e : SUCCESS in 42m 00s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a8a8caeaca114521a5913ba3a16e932e : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff21d6dee0a44f198a5c7a3115699ae1 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b0676e7dc0d4a6f9cdde468d2c0d7d1 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/609924a8cff0483cafbc7a04adf458ec : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"ee783a0ec8b22c2552f5892844ca353f25ad9288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 15:50:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e08c9d4993c347028475819898ba3f4f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f4fa725f17c4bfd9e7bbabd5c9983e6 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff5bfb29c6e4a20b93773ff2c70fe37 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91514c7f1f3b4aba83c0b74074c2f91d : FAILURE in 10m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ecdebe5174447518aeb67c00b5e07af : FAILURE in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8159c0a1fc9d41ed819fc917807cb203 : FAILURE in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e294ad6dd7f461b912a1ca1588f307c : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/36666a295c5f4e298aafbb3bc94fbe64 : SUCCESS in 50m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9523033b399f42cb977d5ab861b99c03 : SUCCESS in 48m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad9faa1e0a3049309b241e92431e51a6 : SUCCESS in 1h 33m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24f010e61a234476aef2ad4b9d3d4265 : SUCCESS in 53m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c90fc09f0415493a8e52d8272326e09b : SUCCESS in 19m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df78cc10ca794ed3b9c75dbb3731fb39 : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/055ea90e9eca4c7684c63717b1362664 : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c1909cd75734a2196e483d87c731e66 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/357b440d15dc4e12b5920a51fd399d23 : FAILURE in 7m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/505887c62aaf4cfb91ef27429aabf565 : FAILURE in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/702dd6a931a0483c8a499b4c345fdfc2 : FAILURE in 17m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7bc9d5ec4eef4dfa887107a9a5de968f : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/899a8772f6ad457e91bcb69780e62cf1 : SUCCESS in 1h 02m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/27995b789f9f42cba95cc6dc8aeb88b3 : SUCCESS in 2h 44m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/982100afd442425490cdffb59a0c0134 : SUCCESS in 2h 02m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/577916e12ae94be3b78561be36a578c8 : SUCCESS in 1h 49m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aea4c3da6e534f628e6718387771020e : SUCCESS in 20m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59f616721dbf44628fdd8ce4124c6c6f : SUCCESS in 32m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/64546275b29d426c8446dbc834baea0a : SUCCESS in 1h 32m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fc5b0cbbbd84f74a9123744778e69e3 : SUCCESS in 38m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/50283d17dff64e0dac1d57a6807ab4d1 : SUCCESS in 1h 39m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f980593b901742bd99d18b437d78156e : SUCCESS in 1h 30m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b730b6958924f3f9b267adbff0ae7ef : SUCCESS in 1h 55m 42s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c4d2bcfd23c49f08d59142a24d6b044 : SUCCESS in 1h 20m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b97df65e74384f6995a99be417b61d2f : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f63bf0430260d9392e380490e87dd4d54eb12ffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-17 06:24:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3da332476ecf3fc365f8cf3076663632f084b8d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 06:24:37.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1721/","accounts_in_message":[],"_revision_number":3},{"id":"1e2457446d77807ade126579de46148861c5e817","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 06:24:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"c799a13c845e6e4d53ee002ead445f737de7f7ce","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-17 06:25:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"8495a63ab985dbabf90d2cd3dd0c7a5ffd0cae90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 06:43:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b95e31ce06b4f6d955ff340a78d1f3a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1d19a67119b4cda8878a9eb38553610 : FAILURE in 14m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81156c05f2474c82bb1598b4625db9be : FAILURE in 15m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9b812f8fd5f40059d839ef5c3b459a4 : FAILURE in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ca6826eec3ddea8f099f78ff8dc3bb8d03faaf9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-17 07:01:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 34m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0beb2084fceaddb6e2e201c3131fd85c2bcf8632","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-17 07:02:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/933216 : FAILURE in 35m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d78112a7b1c9c2e9ea982e7f01a8208da476b6d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-17 07:45:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ba978435ad1d40d68d35bcb5b9af130b : SUCCESS in 1h 19m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b9b8a2f604004109b5d175a7bff50b63 : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d1560cc517901df047fe6cba9ff8dd92fd4810ab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-17 08:03:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9dcd472d084e4e638157381ef820a8d8\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/410d1813bb894786a958cc3e82fdad77 : SUCCESS in 46m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1e5ed9ad8894060bff7236bd96339ad : FAILURE in 1h 38m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fcf1b55b97e465da2baa4c519ab9e09 : FAILURE in 1h 36m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a852e7a418194f18a6721204463edd0a : SUCCESS in 42m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/577a602741dd4ba890ebb108bdd55d49 : RETRY_LIMIT in 4m 08s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84fd9a784fdd41929238ac2414846cb2 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d9bc4bc11dd4338ac9119b62e654bec : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab67189395d040f88281704c3082d297 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJ... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"e765225f6cd66bcab5819d1099a895dca62d5e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:04:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffa511883ea646b796ccf002693e7595\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff39d1cc7c6f4f5489e8e6c11a1a4bf5 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6076ca0e0f374fd19d9829e3e2df1819 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2426f18ac0c5457289959db965d27e19 : FAILURE in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9ed3ba580d442d8aa3009f8af4701bd : FAILURE in 8m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bab899edf8cc48a0bd1f7003c25b3c1e : FAILURE in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925e57caea9440fd8d69af395ac7962e : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0e4cbe594b0e4911aa364df174a8b3fd : SUCCESS in 56m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49a9bb9d586a47e09ef65bbc1a74bee5 : SUCCESS in 32m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6514629f4c6245d3a62b4e1d0b3463af : SUCCESS in 1h 37m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/630bb4e5cc344e5eb31c6138f8102e81 : SUCCESS in 35m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cbd15090bae459fb49930568951dae7 : SUCCESS in 20m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22430ddeffa94cb38cf61ab40da7dcd8 : FAILURE in 24m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d88cd04395c466a8f28088c3aca6644 : SUCCESS in 6m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd8ea0f8ac2545558b5a95e85f8befcb : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee23c47be2784a38a88642b4c3ae0f85 : SUCCESS in 7m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e7cd646adf374bad84b888c7976d870b : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8892f10c4373485795ffa7977c288498 : FAILURE in 34m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a5c81ceb9c944a8a1de2f80b3ede714 : SUCCESS in 3m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3325993287b54038a128f94c314529e5 : SUCCESS in 1h 51m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/94fa375866624954b21a203266fd27ad : SUCCESS in 2h 38m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4031fd97a79f41d48b42b50265f0ebc7 : SUCCESS in 1h 47m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7dd7aab153ce411c9550c4a67ef70d41 : SUCCESS in 1h 42m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/11cf1fcb6fad4317b7194f8be4e2f5b2 : SUCCESS in 28m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f04d5e64da746e3ae92e6a0f1827422 : SUCCESS in 32m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b3476913fe634fc69b95dd54c38a5d52 : SUCCESS in 1h 38m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f60b5351ce5940188160baa089043be8 : SUCCESS in 42m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4734ecfc2d65452588c78b6512fd78d8 : SUCCESS in 1h 17m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9d49e23a9c1d4b38b4acea26b5156e5b : SUCCESS in 1h 34m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/addf9204fe044ba1a9e90ffd9991bfe2 : SUCCESS in 1h 17m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2c239c17b5640119bd42968fc38237a : SUCCESS in 1h 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d141c5beade74a7cb64ff375c80bea47 : SUCCESS in 48m 14s","accounts_in_message":[],"_revision_number":3},{"id":"eaf497efaad9c239f01555be45e0dce55fd044dc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-17 22:06:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/3/20250418023006/NEC-V-FC : SUCCESS in 3h 59m 14s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/3/20250418023006/NEC-V-ISCSI : SUCCESS in 4h 36m 08s","accounts_in_message":[],"_revision_number":3},{"id":"51dd56d2e79ffee6ba81564a9868df3f90ae19ea","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-18 22:12:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/3/20250419011650/NEC-ISCSI : SUCCESS in 5h 27m 01s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/3/20250419011650/NEC-FC : SUCCESS in 5h 55m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6529fd8cae1db38003b94e015a57eaa65ba2e999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-22 09:18:08.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c1ab239e449bdf5a26d2c1385c8a0a9d02cf5544","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 09:48:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf32aa8ddd74031a7f49a7c6f3be12f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3d1e14165ef47848d8024302a6293f6 : FAILURE in 20m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7937a16edec54c4d93d556459a529251 : FAILURE in 24m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/781ff0e27bda485987203b9b9a7ce15c : FAILURE in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4dc5c5be7045ee148683b03f90471045c084ecc4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-22 10:00:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/933216 : FAILURE in 40m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f2957b494b34ca50b8de8a4b6cac42edfcfc507f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-22 10:04:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/933216 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":4},{"id":"872ec09c20cc59d6a6081a5463d36a2d244e14fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-22 10:20:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":4},{"id":"f18594e53b6f6c90976af620273f81895421cdfa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-22 10:36:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/40d9c5b1b93240fab2a197a3b6e392c5 : SUCCESS in 1h 15m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7f4f4deb5e7c413a98a272d779360441 : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8df488934cdf8194dd62d9254e305270b024b1b9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-22 11:10:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/67c3c9c37f4e426f997c2b712f1c06d4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/183ce3820fce49e5b7d5e88826c98f91 : SUCCESS in 42m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6f23b41972f54cd6bf852d5dd1fcf857 : SUCCESS in 44m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/06885b0c384c4d62b55fc1e247580164 : FAILURE in 1h 51m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c2b85f39cb941cc85fa18a1e77e83f1 : SUCCESS in 40m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e81aa481e4f45dcb600053926ab4156 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5659d662f0b4d37a68a56d0b745de1d : NODE_FAILURE Node request 200-0000039112 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4ab5daa77104c0bb36a8597e735bd36 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f287159b67de4a8f8536e5da49387719 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"bdbcff610b386d158d404d8387a8fca3e8a9150a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:33:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e8fe56d52e0479dbcd62b4d94a37e95\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6351f0684997423aa39b9443dacf86e5 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383534974d4049828bda331a62b6d5e1 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7d055b218aa41fa95310b272fe6bd1e : FAILURE in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c153191fe9d94cb099ab09ee382d283d : FAILURE in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4df39c80db5245d18e0b523aaf1b930a : FAILURE in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35868e02816459c94294974053d7d8c : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/09479bd42027469eb9e21658e4cdab0e : SUCCESS in 59m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca356103c200485eb2e97fc9b44b1b61 : SUCCESS in 42m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef4ffc39d1814a079ec29038627f9186 : SUCCESS in 1h 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81a9ec92bab54f3b8f87cb659cee2d47 : SUCCESS in 50m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adbe62f95e6146be8d39c3ac4d301445 : SUCCESS in 16m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7640c4d367cf47179b7ddf41fd0f65f1 : FAILURE in 23m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8b998f130834438db764e38c8ed2e7bc : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/314a0335de994c6f9831e17f48200fa7 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85edff1d8c874c80888391397f50bccb : SUCCESS in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/892a46d43fbf4a8e8233705086a35c70 : SUCCESS in 8m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef48c54b318f4950bda6da4a49c47ad6 : FAILURE in 29m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5c7e0dbb0314bf8a953eb30f1a9fe7c : SUCCESS in 4m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/510dbe0affa44f29b489b382cc0a9a79 : SUCCESS in 1h 55m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3eeefe8a6090479fa453c10b9e50b1e3 : FAILURE in 1h 35m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/809c9600b50a4d88a236921ffda06005 : SUCCESS in 1h 39m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/449cfa94d33749c78d3a37f25bb04b79 : SUCCESS in 2h 10m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4e3a22e6607a4a05bcb1da5c0973f202 : SUCCESS in 33m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6bbf1dd914e94760888b5fea000c9e99 : SUCCESS in 1h 04m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c15b61fef4a348fb80ef53d097e64a2c : SUCCESS in 1h 26m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d437bc1fd244438b2e09330535ac209 : SUCCESS in 42m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6a61eb5ba38d4300b32d2659a46813e9 : SUCCESS in 1h 20m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/61b6a46fc9af4b16bd3d9ad3909547a0 : SUCCESS in 1h 56m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc037a67f8cc4e29aed2850da349c107 : SUCCESS in 1h 29m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/677e9dc3908547ab8b428acd95f2fa39 : SUCCESS in 1h 21m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a593a96d4ae450c85d588fd1801c899 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":4},{"id":"50e15b88e306d0d2a77fd262b61e1a6ef7ceae5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-22 12:56:34.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2a6218a64c28ef966618a3e734ed9ecf9b43d8d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 13:21:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8234ebdb474947b4936e5dc62e9b3560\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ac7b6ea7a614f2096ca967bbf4563a6 : FAILURE in 20m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a87e5853a4394d9e88bda8c123e99991 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd9c9f2bab894167a7cbe838f5151423 : SUCCESS in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8eb8c3d3738ff53175d957132fd868c2bf8cfba9","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-22 13:48:05.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/933216 : FAILURE in 39m 52s","accounts_in_message":[],"_revision_number":5},{"id":"5484913dde3e5e50cabe1f37a2f92713991bd659","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-22 13:58:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/933216 : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b9511a9843fd8f5fac798bfa4d58cfbb49935216","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 14:03:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/745/.","accounts_in_message":[],"_revision_number":5},{"id":"64a7f8a5a9eae570eb97549d4009120ea836213f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 14:07:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/626/.","accounts_in_message":[],"_revision_number":5},{"id":"1d4dfb87a41bb9ec078a00132b83fcd50083dde6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-22 14:09:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":5},{"id":"e738f4ed23c115cad9a9669f853ca7d4d3bfd576","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-22 14:16:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7bd74f66ee684cd7aafa3eafcd642da5 : SUCCESS in 1h 10m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5b2597b4cb9145428b9ab5595fe9ef48 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":5},{"id":"0e597d44dd0812e689f9d6eeda48cbb1de8bda41","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-22 14:29:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/40eb0b1f42014995948be51d407ada9d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9ef78a7c7d7406398b5277b355effe5 : SUCCESS in 39m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c51d9d2ab85b4f6b859c85ba8a84ba5f : SUCCESS in 38m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/247a3d31b4f84f6a980a97f34900f894 : FAILURE in 1h 31m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c26db70df8c249eebf64f481acd29ed3 : SUCCESS in 43m 19s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8b6bac59bd942898b1c318210897fbf : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6357b74c5a08429fb1f1c3fb59655765 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6b2fcde7f814cb8941cf293ab45c3a5 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab326fe7e8234b5e8b04c5d60ed983bc : RETRY_LIMIT in 30s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"6c52282623af1ba21edac42e5d46bc3b12a4dd89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 15:02:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6d9829c86bf4b8aab4169ba4d6e0cd1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b42dea174914162abb554c8ef16155f : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ac08e389ff4b0f990f823c00c23d6b : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/592e177040ed48dc8d8948fed69f78ed : SUCCESS in 9m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3898d73ba7804a0b8189d12f38c5eacd : SUCCESS in 8m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9616f9f1080042f385517e82bba9db1c : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/449748d35bf5478da68ba2c380d09b8e : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/004cd8ec0a5a46deafeb5a25f454bc00 : SUCCESS in 33m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c89c4eedb01d4dfba9e95f930a215070 : SUCCESS in 29m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d8569333f324dce8aad054f9c811bfd : FAILURE in 1h 30m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82fcdc42382849cc9634f75baf3b5988 : SUCCESS in 32m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf100cc19a64d349da372d6374d9810 : SUCCESS in 11m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/946ae614b8174c9895df1f58165b0618 : SUCCESS in 17m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4492231dc2a741508816caece798af74 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86cbcc76bd364790a97143701524f3de : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4e663ede185448b8495f852a778131a : SUCCESS in 7m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b49f64482e35477189f6a8588779547b : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cefa312a47114de6b685033e437ca6cc : FAILURE in 32m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/958a7c0eee9449efb412168e7e54486f : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bdb099d32ea04de09ee6268a1d266149 : SUCCESS in 1h 35m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab98a8a322f84a148764f0a08cc21ab2 : SUCCESS in 1h 32m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cdb8de28ab5047a093566311a7fbbcc5 : SUCCESS in 1h 44m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6fe50988ea2546ad8a23ae68509d4218 : SUCCESS in 2h 04m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a940d6039eed4eabaa9373b77dde8bfb : SUCCESS in 32m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/28d7b3f414f64f75ae15d5643306dd80 : SUCCESS in 57m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/759d20e1c1844a7da8ef5eda6484f881 : SUCCESS in 1h 26m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2abe79b174e7412a83c814ad343dda8a : SUCCESS in 44m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e945d261bc244f1994ad416829590a83 : SUCCESS in 1h 18m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c55786ab19034f1a8e1b80259b81a833 : SUCCESS in 1h 52m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22515530655044f18e1c3a619eb48b4b : SUCCESS in 1h 12m 50s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c04cd7ee274542d2a8ee1514b58d8fa0 : SUCCESS in 1h 23m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13ead71b99ac4f138cd6f29ec47febad : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":5},{"id":"c2ecf6f42cc87a57d0d3425f45d689dc492285b8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-22 15:28:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8bb4a5f671ad881786626f956f423b7681a1c34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:19:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e2f3bf1b1704410bee7a781792c5580\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b4f62d6ffe44b6db0892bd9d63294a1 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32e605db91374f8bbe97309254c92dad : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/997371ebefa3442481125c9f01e44227 : SUCCESS in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0c72236258331647ed54ad903b9a6b36e17fd6ae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-22 16:25:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6bbf802ee1384ec6aeffe0578a7ef332 : RETRY_LIMIT in 14m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/524db6fcd21e460c980271ec95d6992c : RETRY_LIMIT in 16m 58s","accounts_in_message":[],"_revision_number":5},{"id":"d9a0158feef7289d44ea8e8f1be6eeeb7db7507c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-22 17:10:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/4/20250422213618/NEC-V-FC : SUCCESS in 3h 55m 48s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/4/20250422213618/NEC-V-ISCSI : SUCCESS in 4h 34m 21s","accounts_in_message":[],"_revision_number":4},{"id":"efc5498c63aed786a3a320c78d776bfcbbe686b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:44:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44035c745a1548a9adff75791d856b44\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b84dcccec53542d6bd9c2be080cc95d7 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc8e467b20f453dbb0965bda87df893 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd22740dd29d4c9491b3a3636838a6a1 : SUCCESS in 11m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97ffb74829094c9a9d40ba856fbab7b8 : SUCCESS in 7m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdaa0c0230b34c93b7a11794e7ab778a : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4400e0b8fe7489aa01313557b428164 : SUCCESS in 15m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3ed6bff626c64fe48dc999e306e5cb33 : SUCCESS in 51m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37d2b12e9b1846acba9e3892f4d6ca91 : SUCCESS in 51m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47e4e1e401294d04a4e9b0899ccd9eb4 : SUCCESS in 1h 26m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eea33ba5b8a7496f8afad0de6b1be6d6 : SUCCESS in 50m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4068c05f5a4cf2a200d34e843742e1 : SUCCESS in 15m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7601eb18b2149e4a729220ae74e882f : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7e995506531d48c6889394a19a3fd147 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/624eb5e0385142dab3d8d98803874cf8 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7901616d4faa44a796418ad2837bc662 : SUCCESS in 6m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b0f7d927f5ce401c9f8d7c604c6c6257 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/34355c26184941119db11e8fc8fb6fb9 : FAILURE in 26m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7acc0ee25b240fead048d58f76112f4 : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0a3dbacbf804a32ac13ae1fe089104d : SUCCESS in 1h 03m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb42e6842abf444b88093f0c96f6981f : SUCCESS in 2h 14m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/194082d1c2ef447fbcd3319cdf96c6d0 : SUCCESS in 1h 28m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/766ca8014ea14d119c1b81d853477d6c : SUCCESS in 1h 37m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/975bc2a6e9564eee84d89e6480cbe0a5 : SUCCESS in 34m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40f76534e4fd4f80b9f30efc7252b82e : SUCCESS in 54m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/515c35c824664f70ba1af9943efd8ed3 : SUCCESS in 1h 30m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4dcb1e9f45af4349b555f63dbd502f95 : SUCCESS in 38m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78a8193cfcaa49b888dc06b6fe7625d7 : SUCCESS in 1h 23m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cdf7b01779174f9083903cc27005818b : SUCCESS in 1h 27m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cb06a6d3ecf42b49474a7060c36c1f9 : SUCCESS in 1h 29m 06s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95acd12afe74445891f108f5d6a6d5d2 : SUCCESS in 1h 28m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f962dede64344e6c94a800c42f55626d : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":5},{"id":"1d625ac17998c285dd00dbf8c76cfb2750b20e0e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-22 18:51:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"883c762ec94b501c648b02fd062188389fa0f613","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-22 21:16:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/5/20250423021046/NEC-V-ISCSI : SUCCESS in 4h 05m 29s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/5/20250423021046/NEC-V-FC : SUCCESS in 4h 05m 35s","accounts_in_message":[],"_revision_number":5},{"id":"0f8ebd100cc71317f1263b7877aca74d351ff0a1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-23 07:08:57.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/5/20250423061311/NEC-ISCSI : FAILURE in 9h 26m 30s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/5/20250423061311/NEC-FC : FAILURE in 9h 55m 43s","accounts_in_message":[],"_revision_number":5},{"id":"7ba71308a00a5e3528c28378321f317cc165245b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-23 09:28:59.000000000","message":"Topic set to bug-2107469","accounts_in_message":[],"_revision_number":5},{"id":"e68376dddb783ac8e13b1318dab71032034fd26e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-23 18:34:35.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ee1a332788f9d9de37c41d6c6399e58d701ef9a7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-24 08:38:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/933216-5/HPMSA-iSCSI/ : SUCCESS in 5:44:22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/933216-5/HPMSA-FC/ : SUCCESS in 2:06:59","accounts_in_message":[],"_revision_number":5},{"id":"dee8f50343dfe2b4b0ec5a97ee4979ef0cb25c53","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 09:50:13.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"8537c2125cf3fb1eca18c60afa7112f52a13a313","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 12:22:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"db12b2e9b5fa44b62a646a593755de8dce043070","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-24 16:51:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/933216-5/PVME-iSCSI/ : SUCCESS in 5:42:59\n- PVME-FC https://os-logs.seagate.com/dell-ci/933216-5/PVME-FC/ : SUCCESS in 2:06:56","accounts_in_message":[],"_revision_number":5},{"id":"939aa16907f62f03c0e59402c8de40bf3f087b71","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-25 00:58:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/933216-5/Lenovo-iSCSI/ : FAILURE in 5:38:03\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/933216-5/Lenovo-FC/ : SUCCESS in 2:07:39","accounts_in_message":[],"_revision_number":5},{"id":"a65d474649cc69f1b40637bc9a3f66f12d89e4de","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-25 09:10:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/933216-5/Seagate-iSCSI/ : SUCCESS in 5:43:52\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/933216-5/Seagate-FC/ : SUCCESS in 2:06:55","accounts_in_message":[],"_revision_number":5},{"id":"032a63f9d486f8da2ce1b33a5a41ab604e4abfc6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 09:42:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"68e22155e0305e62328d9014bf92d3ab14d53137","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 09:43:32.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"05b02571b3a0620a136e09a15f97bc25268b3aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-29 07:09:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"aeea640f35e528aac09e8343c258a7e9f2daae6f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-29 07:11:19.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"4cefa981f77cd77507278b009cd97e253d073495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:02:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/def2df66e85a479ba5967bf32b990ac8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/133eb48ed7214acc898e84d70011a378 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/738aca557cec4b2ab4ae28ceb5b9fd7e : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/71f09180639045da94ad823e87816692 : FAILURE in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"64509faf0fc3dfa71b6ed5b31ad01527a9f393d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 09:50:22.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/354/","accounts_in_message":[],"_revision_number":6},{"id":"351a0f7dfa7309f2f8d7e46c070bdbc6c440c85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:25:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998d2298ae494bb1adadef1ea338dccb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec0340684a9946b68b234793c2ab0392 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791738cef80543288eb83ccc03ea8b72 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4eb97a52e81c409e829bdad074b7f8d5 : SUCCESS in 7m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26af85eae838415e9967ed2ef1f09708 : SUCCESS in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e270a05d97745e88585cbc056c398db : FAILURE in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43525f2c69844f5caca0eebc3493d274 : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f34c1a34c0354d139d2c556e2fbd7b71 : SUCCESS in 1h 02m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/981fb87bed0f4605bdccd9f1091fc9df : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ef582f8e47b47fabad0f96ba8894e3b : SUCCESS in 1h 30m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c7b795fd1c14c29965e0165a3b29b79 : SUCCESS in 32m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04dd038a3b4d46f8bcc8739e8e57dc84 : SUCCESS in 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/88a257a3edcb404ca00133ff7982e1a7 : FAILURE in 23m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4c26ac6bd24644d8a996a9e2ffcd9c40 : SUCCESS in 3m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7536ca265bb14797997de9060162dea4 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91491e4e53db4e38b202db0bd7d732ed : SUCCESS in 4m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca579f09311c445e85031a2852dcfdd9 : SUCCESS in 10m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4f796c6333e2447e98ed40e5e517e08e : FAILURE in 18m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22569205d5ad4ed0876cbdb9ffcda5ce : SUCCESS in 3m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84d944d6226c4ae4ac03a21f73ebc2ef : SUCCESS in 1h 11m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f71181a1c067495eb14097c4f0fb9099 : FAILURE in 3h 03m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9928705cafad4c0e959ec29d011e435d : SUCCESS in 1h 53m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/50dd10caef764599abc6b5aff13cee03 : FAILURE in 19m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6252695f777848d8a1b42657895e7d05 : SUCCESS in 17m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de8c96643ab74200836ad6282af292fb : SUCCESS in 1h 08m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/279d5a7d2fc7478898ec8d5faf6c97f5 : SUCCESS in 1h 28m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d52488a6e113493fb791d203f04bc706 : SUCCESS in 34m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c38a48242954417d901a8159cec53c85 : SUCCESS in 1h 21m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9a8f30adddbb4ec892ec783ae2f7510e : FAILURE in 11m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb2c7aaaec4c43a4883f6db4b6e2916c : SUCCESS in 1h 33m 21s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781edb845ff043d39945de8828095bd9 : SUCCESS in 1h 34m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e51e694ba1714f3a9b012fc3efe25f1e : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":6},{"id":"71571e57440d722abf322e2debe67b5991058718","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 10:43:54.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933216/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"3ed672d6123fb7c67cc03debe02f39bda23801b4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-29 10:50:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 36m 51s","accounts_in_message":[],"_revision_number":6},{"id":"211e915a60ecabd33f7c3d99a3aa41c2d6700841","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-29 10:51:02.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"848a7c8ed1d699269f1b666702ab70615a0bf128","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 11:02:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1613/.","accounts_in_message":[],"_revision_number":6},{"id":"c8c79f705bb4c7ce3856f1274dc5e905f5ecccb2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 11:11:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1747/.","accounts_in_message":[],"_revision_number":6},{"id":"d2cb99c17a1d9ca54786e181707e2b5669669391","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-29 11:20:29.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-933216-6/FC/ : UNSTABLE in 6m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-933216-6/iSCSI/ : FAILURE in 4h 6m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"6a77c7fd6cd211b4d58d03ceb4cb53d9f00bb33d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-29 18:59:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/72f7e3c718fe4b2eb83753e96db49415 : SUCCESS in 56m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/adac8ee6b9124e849488eabc76b6cb3b : FAILURE in 1h 01m 00s","accounts_in_message":[],"_revision_number":6},{"id":"d0a6c188f8c127638ff10248d20b5f89125c29e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-23 19:11:41.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2de6227415d71f9a6b683c9801342419812d68a1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-23 19:13:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"ff99ce504b2fa5b1bd8a23ca8d7c726ecaf7a0cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 19:34:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d8ea1ae08243069d7b22fc19cd675d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ce3335809884d888f06ea9d99c3f194 : FAILURE in 20m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cfcea5817b0c484b90ae4a42dbbb1279 : FAILURE in 20m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/731bc38f1c2c4a23b63a0c8b19bb2b30 : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5df94aa724782263beddcee44823302f128d64cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-23 20:07:55.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":7},{"id":"29e8ca92c4fe25ef39bd86dbe73bc49eacdea8f5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-23 20:29:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 36m 15s","accounts_in_message":[],"_revision_number":7},{"id":"18cedfc8e05b4f9493451a81368d3b0b86ed2d44","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-23 20:52:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ba16e6c3e3074efa99392748c6554964 : RETRY_LIMIT in 14m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aff43125161f43f4aea5f08cd9bd7517 : RETRY_LIMIT in 15m 34s","accounts_in_message":[],"_revision_number":7},{"id":"a1cd2bc97b8172346f51e494d3165c199f1217bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 21:14:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2121/.","accounts_in_message":[],"_revision_number":7},{"id":"30d00140df54e1276c7b1d839cebe6143b77c11e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 21:17:33.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2235/.","accounts_in_message":[],"_revision_number":7},{"id":"21aeaef089770bf5f143a400931e315278a2491b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 21:27:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3aafb379e9745098f45fe8371d2eede\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cc5ac26cc6f7455d8aac68c68174ea90 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b01f331fbae496e8bc2f4896022b492 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a81531722a60455c92e0e31bb0b2c265 : FAILURE in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2339dc41ce0341d89c9384f3021f84f9 : FAILURE in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b3fda3bfb44ec1a397cd7256d8daf0 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/eb310f6cd4ca46dbbdadd81cdbda0839 : SUCCESS in 1h 00m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/624aad1c5e1b449e9a4b527d0fff3d36 : SUCCESS in 41m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/903c4b01009844caa88ee82aeb1669cf : SUCCESS in 1h 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/238512c43f7f4928a8439024d0e14367 : SUCCESS in 57m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9a26160936443599e036b7861f15081 : SUCCESS in 12m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d51f56697f34431c8027634be4e1aac3 : SUCCESS in 16m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dac05d69bb9443e6bcfb8122aa669ec6 : SUCCESS in 8m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c79df2e9f83409a80b8854e84bd30a4 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/245314f0159642be9c2434c103e1cf0d : SUCCESS in 6m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/51c6547886f448bc994d25129d949726 : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f61da44839c24483abefc6f75eab7c30 : FAILURE in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9d5df3eccba444dbf0633cee5fafed4 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/618d82f561be4621a9c4bfbff1f84178 : SUCCESS in 1h 03m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2cb7a1f83af5408c82433ab46c5fe694 : FAILURE in 2h 11m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff105d886ff54f309c1a343c30abfc3b : SUCCESS in 2h 12m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/82c32102f5bb4be98cbf9205e273353c : SUCCESS in 1h 52m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ee432b9bb006467d9ad02f537a231fa4 : SUCCESS in 32m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0dffdd458ac540108ae007acc1d4af86 : SUCCESS in 1h 09m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b1f335c968ac4d45a9d780930028f268 : SUCCESS in 1h 10m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1de80482bc574e48affee2b81ca217eb : SUCCESS in 25m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/66fcd8eb5ba64f3593942cc514721e29 : SUCCESS in 2h 10m 25s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ee4b92980bc44e25acc06faa63500c46 : SUCCESS in 1h 22m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/caafb6d17ed24e698e1bb5019fe04eb9 : SUCCESS in 1h 37m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06a39d7680c24ee7b58a2599e66b8482 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":7},{"id":"28febfb13f4329b03b7fb4ce9bccfb96af4eaec0","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-24 06:06:25.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6606d88dcd813550af004ef0c4d143d888f627e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-24 06:31:54.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8f0db3fc15314ba36352f94039bd115f43cdf3d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 06:33:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"cc03416ae33f6394c027bf85ce8014efd729bc0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 06:59:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448403e90c5e4b2fa6283b7ebfa5eb23\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8232b896991747889e35761e72b85b33 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25c303d5c27d4ca4829a8769a45cc312 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43654525f7b7483a9ecf73747cbf47d7 : FAILURE in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f915b30b0527d09bb69e0501ecc8dc9dd4a2ebed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 07:28:31.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":8},{"id":"357eed29f9954565026e5399df8ee3f79be0d263","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 07:32:17.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2239/.","accounts_in_message":[],"_revision_number":8},{"id":"4e847f414cd8e2d2c1ffe6d76c0e13aae15684af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-24 07:40:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1129c199ffbc4a6ab6b56575d1acd564 : RETRY_LIMIT in 16m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/95faeef6583e4454b8430706e5cb5b2b : RETRY_LIMIT in 13m 42s","accounts_in_message":[],"_revision_number":8},{"id":"44dc5e6daf71c5502c2b25fa0c36048e48b67bc9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-24 08:19:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":8},{"id":"d3c2b7b13246cf90d418c8b93ee4117c45fcb514","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-24 09:03:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2125/.","accounts_in_message":[],"_revision_number":8},{"id":"ea935c73ac5810952a47cecb40eaa17ebffb2e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 09:41:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39517791077a4d239522aa55195365a4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/63663c3141804428aec2cce74948b076 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99991c4b219b4a53bc8e98de10d01ebe : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a054fd70da7463fba58919bc48a7a21 : SUCCESS in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2052988271f34f03a7f62176ad834530 : SUCCESS in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2227658484aa476087e5c3cf001e725b : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4a82e65996f54f94a35241775b4ccdaf : SUCCESS in 47m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44b410a87fb34d44868e62840f1c08ba : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02d83ffbd8514a729809b164a62bcf07 : SUCCESS in 1h 18m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0390c0c2b34e484fbf237421c2ffbcf3 : SUCCESS in 49m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4207fb3643834aecb2a20b57a67091a6 : SUCCESS in 16m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/736f44d36143407d9380bc39912ceff0 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f4dbf1e1d4e3475ca37fd668204559a6 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4144bb5496e3486fa3006854c085991b : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0973cae498da4db2bd6a659fcebcf45b : SUCCESS in 6m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/31d220b895f245ebbacd5366e93c861e : SUCCESS in 8m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3be743e254c74320ae4d128dc45bed30 : FAILURE in 36m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b467a79e9065483a9e1be8199fac506f : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a83e5c0d658a46a4bc68bc4b43e3639d : SUCCESS in 1h 52m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b70478c8bb9448de8be47f6b1abbd7c2 : FAILURE in 2h 22m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08568528147e4af0bf8e2c1e34c3234a : SUCCESS in 1h 47m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aa3fd0fb440147469a76e4f9b7d85991 : SUCCESS in 1h 55m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4027efd72fc4d53be798b9582ee8ea7 : SUCCESS in 36m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f00d45215b58425c81d52863daeb55d2 : SUCCESS in 1h 07m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b83bb19ff2e4ccabf085805dbd19781 : SUCCESS in 1h 22m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/58808c3ee2dc4d97bc57cd0ae4e71cfd : SUCCESS in 24m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c8d0b18dff443fb89f3ce5d01b28ee0 : SUCCESS in 1h 36m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/dbb2805120684ec1ab9ed60bb72b0668 : POST_FAILURE in 2h 50m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2632c52dacff49398b75d8787b175dba : SUCCESS in 1h 38m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5451d3715ea14673ae2aed5964bd81a0 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":8},{"id":"2247f43922fcf2a66f571ef94f0acda7b9509e8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 09:49:02.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_09_42_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"ff20ecd941e6e159c05b4f20ffd9da4a585d0df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-24 13:58:09.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3e41067b2bb154ac69e25694c3d5e6f4fda346cc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:59:19.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"6b823d020dad6b139c2b47a598283e3d4ec28065","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-24 14:53:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1306696d2a87323cce7b98a3a9e74b8e5e55d753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:57:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1bbc20dc9564e099c874a76bb2cfc1c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bbab62a0344a427a87cab7143d0072b4 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca8a989264941c19d6877d7f22478a8 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79d525096a45470196e6ef623239de9f : SUCCESS in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0222c9fce524dd0b0eb7cd366ebc7fd : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24663eb8be1e4ed1a563e1cd63cd3dab : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/bf46b11fae8a4701a3586fce2540ac5d : SUCCESS in 1h 01m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c28ab902128f47f0821cc499b91aec21 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea80a6c4500e44a699eee0825a84510f : SUCCESS in 1h 37m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/271de998df6d4ca2af0f93274449090e : SUCCESS in 50m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79b6daeee31344df96a65bfb9518c3a8 : SUCCESS in 18m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f8f2e5e7d284183ac9b037501b60745 : FAILURE in 23m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/796c32b3c8ff41cca1d30c836ed44529 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfc4304b4a314cdd81199ca6d09b0b66 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87d1581d60004f36954aecc83e67ae60 : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2694e6e0e09d4cf8a935a859edcd135a : SUCCESS in 10m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a03fd4882c54be09d2d2d1bf6eea149 : FAILURE in 36m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d0ddd801c5645088a63bbe3fa6f393d : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58c31bed5d8b484c96f869b695d0f657 : SUCCESS in 1h 43m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/844bd8ee64d349028b1387cc49c0a69b : FAILURE in 2h 53m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/708580008fa04da49236d8ac79fec42d : SUCCESS in 1h 30m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/66851db5e6934012856be9e4fc026a9f : FAILURE in 2h 01m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/89e1ee5933734b1e9898866cb1857615 : SUCCESS in 18m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10222f0cfb2147ac8e679d12fa92854c : SUCCESS in 1h 04m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/273058edc077456da08779873cc9bb6d : SUCCESS in 1h 26m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4fbadf671b114a3c8d31ba47ad762697 : SUCCESS in 38m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ea8ede8a7b14637bbdc03f1ebafa471 : SUCCESS in 1h 39m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/124648e29a8e4f3fbb50af970442f362 : SUCCESS in 1h 58m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5776d49f4e6742d381146c11fc195a38 : SUCCESS in 1h 34m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43d4dcfba8424eb58369b1581bcb7e7f : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":9},{"id":"4cefd2270c4c1e042850f034ae4fc65c48cedea4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:01:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9470cc06b6ee458bada0cc45380122eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28e01c83bfbb45b1b2fe83fa15265f2f : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95de1dbe5e8a41aea6d0cbd10dd1d0d4 : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9f98a3a15034444a16c6f9e59cea6be : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0ab72bd7ebea1fd9e36daa156dfd219e2bd75184","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 17:04:23.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_16_57_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"bdb7dc43c2e6dd43d3b1cc45d6a81803ad30c160","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 17:50:39.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":9},{"id":"1ded116795701442bdd417a415206b267df753a7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-25 03:46:03.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d62a6059f1378be39a9dfd653bebc1d87af094e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 10:30:20.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8b728eefd47845b5b296965e11df94c8 : SUCCESS in 45m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/01583c2858bf469f95940f688bf92938 : FAILURE in 1h 04m 14s","accounts_in_message":[],"_revision_number":9},{"id":"3226834975f2a1881c8fa5cbc4a250878844629b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 13:27:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 35m 58s","accounts_in_message":[],"_revision_number":9},{"id":"de78ce2f3a8b7ba3ed226c61277e5e7001fc49d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 02:37:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2171/.","accounts_in_message":[],"_revision_number":9},{"id":"17543121aa37621cc623baf15afec659a0ea3a2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 02:50:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2285/.","accounts_in_message":[],"_revision_number":9},{"id":"fb5f69889b127ed1e5dc6acef400ae053a61edc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-03 15:57:29.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b13af499eedbc5ff918172b34f529dbe9f6ea33b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 15:58:59.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"d1a29c66200a69101edfd132005e440f15bfef51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:34:22.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b4e3b2c20c4d9cbe3f9e1907ce3608\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d72287e8d0b74215a639288d4aa7e112 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e0abe791285444da081ae4e38171b8b : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2dc0e67e955640f7ac208bbdb706f855 : SUCCESS in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0cfd0313c5735e84a3df7337716408f7262ea283","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 16:55:15.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":10},{"id":"d1a6d02dfb9c499b1c55162d32348ee78bf37083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:35:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0931caf2d3184b3b8e1b74b02c75becd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e1be4ab79b345cf830068823e2466d6 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1498b9bfeb488a99f90c3d6bfd271c : FAILURE in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c479b1d6eff4b49a6c4ba44cce8717e : SUCCESS in 11m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44fdbd23c47c4f3f82ddce1f9b8e3c8c : SUCCESS in 15m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf77ab83fff46a19e5b84756efee1f2 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/99e078c1db3b4b35a596bcfc8da68b6c : SUCCESS in 45m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c5ef7aade4b47e5be4fe00f6830604a : SUCCESS in 57m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1220f692f4b45b8b680f4fb862be03a : SUCCESS in 1h 36m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab5baed71bef4c2a9641e8d3b4fc6876 : SUCCESS in 59m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/397be58549174ffe8c4191d65c3b9c4e : SUCCESS in 15m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c42e344a999944a8af7676a4c1835335 : SUCCESS in 19m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5bc28d9f6ef4873a9662ff902889a1d : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/64040579d1314b6783765ea55e5a5e23 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b0c0409df1e418caa3383f98171ce14 : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f81c6bfa39434612bfc04ab27114004d : SUCCESS in 9m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3df5d13d7616417ba7866a11f55677dd : FAILURE in 19m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2df2b8871f164f0593db014b2f907423 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10b7733eaef14f549a71e789ce9c1c10 : SUCCESS in 1h 53m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41c8b6bfd75a443485794ba6d7585fe0 : FAILURE in 2h 27m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b6a36452c4244bf4b8d6f2fd2845dd3b : SUCCESS in 2h 04m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/41f2b9115c2c42ec87e5ff9c556946f2 : SUCCESS in 1h 44m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b3c8c2058d6640edb5abf3c77e9fd72d : SUCCESS in 28m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59b6a8f69e54412384527398810dd36b : SUCCESS in 1h 08m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9d242435eed431e9c4c46d49f804047 : SUCCESS in 1h 51m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c10375a8016b43c58f5fd481d5f5b421 : SUCCESS in 34m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/97ad33218d464220a51e1b057ea842a9 : SUCCESS in 2h 01m 10s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/72ae4142501c4a4e909e7fc8fe067dd6 : SUCCESS in 1h 13m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9477a6d3e5f344db9bfb39962668d53b : SUCCESS in 1h 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3b95332d13e487db409aeee7d046719 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":10},{"id":"f0a713a8d73cedb6e4361915cf0844c5d9fb7c22","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-03 19:07:31.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 36m 40s","accounts_in_message":[],"_revision_number":10},{"id":"8d3710580375e8cf840d841da7adab1519849dac","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-03 22:37:46.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0e6dd42a62f54ac8a10e2fbe78246ce5 : SUCCESS in 1h 37m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eeefb1a4238c43d3b94281c72c9de9de : FAILURE in 1h 58m 33s","accounts_in_message":[],"_revision_number":10},{"id":"89b33425ac0773ca0c8e9184345597e92ca6a23e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-04 10:42:55.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"152b6370693965fe185193beeb61fdd4645757b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-04 10:44:02.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"568c14a3e4ce69c8850af26d97e9eb4080356213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 12:02:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66644eda66054258aed118556e180c9b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89c3759471c0438c8c3c68f0af0be04d : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c100962a66384cee9014a7343ba7c181 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db9e7dacbd3d46e6b1ff85745da02fee : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"011d256e83dc56cd914a415aa238b198050839c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-04 12:47:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 1h 11m 30s","accounts_in_message":[],"_revision_number":11},{"id":"a4198b6d37a0a24806e33a0938517483762d0359","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-04 12:59:11.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/653ebf58370e447ab1fba0100027a955 : SUCCESS in 49m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b471368421ba48af94dd8e75d5b6821a : FAILURE in 58m 35s","accounts_in_message":[],"_revision_number":11},{"id":"78b0a29495c12de4c8aa39b9fab04672aab283c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 13:39:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e59fc3898046cfbfa936aa4cfcf11d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e8db4d06809b463b987757b3f224f593 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9f7289daed40689d8e1c81b5b82477 : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bc3185badaa4f6f8ba416312e1433fb : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f80893822bd34ccb98ca8a3ab5e21b9a : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ac2e53364841879a44a27a7121dd66 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/a23fdb1ccefc4abdb520ecc81f29fe5d : SUCCESS in 44m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1601e25742d9454ab64acb31f1373366 : SUCCESS in 44m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b5ea18aec6f415eabebd03629d1dad0 : SUCCESS in 1h 26m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/108eb0e8f1354e1a809fb61a52e4a05f : SUCCESS in 48m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a548b270ea341a39cb9109b916bc992 : SUCCESS in 12m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f60222187514a90aca58a16c1092436 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/750cce654892494ea5b30907c0cd1994 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9745c65e09f442589f068db564273591 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e00b7e7eb304bb4b2487fc9da81adaf : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/769cb2f10eaa454c9c8ca98942660010 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/54eda01658684cdfb092414dfb46706e : FAILURE in 39m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e0265eaef0f495a99312fc4f8452b59 : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/091cd38275db458eb9ab6bbbb69512c9 : SUCCESS in 1h 50m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9dd14a8a0dac47efa592d0a53c255530 : FAILURE in 2h 44m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ccb4ff43790479cb11ab1333a53f80e : SUCCESS in 1h 50m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/51f39ced5142445db1b87f177255e56c : SUCCESS in 1h 53m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f70611949ff9481bbfa5c3a08aee4df8 : SUCCESS in 25m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1cf45376e8964d609bc01ba442b8952b : SUCCESS in 1h 07m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4476bf515e8c4a4a83744901df50aaca : SUCCESS in 1h 32m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39513428431542beabc90856be1ec28e : SUCCESS in 49m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b59691a0d3f4f89985862d8cdd55786 : SUCCESS in 2h 06m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7b374b8242674ce0a3911844c00a2a38 : SUCCESS in 1h 54m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/319129ad779748378122c4cbc3c564d0 : SUCCESS in 1h 34m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aaf51c9a2e94125ad38941cee53b463 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":11},{"id":"5039c6a4f0c26375d6788eaad3ac791f1122fec4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-05 10:47:11.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2412/.","accounts_in_message":[],"_revision_number":11},{"id":"35fc42c75e4155ad8c7cc0c59298a9bca5e5ef8d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-05 12:29:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2298/.","accounts_in_message":[],"_revision_number":11},{"id":"09b2a60b01c7527548db27d2b2a6585b1cb3d784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-12-16 15:32:30.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"13ca442712bf7f6ae09a7c517523d9ed6d416b76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 15:57:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5ebdd47ced4e22beb523a60f8ee3a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f7a80e5c24f44bc8ec3323d8a8ad7cc : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/966da761009c44b29c04ceb518a24984 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c386ddaa270447f8588a119689cbcc2 : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1f0994cfd2d3e3d15758be7fdc24d1322b95f4ff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-16 16:37:08.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":12},{"id":"d36b2c8988cfe123d4b0048cead5c02f6494371c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 16:42:27.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2756/.","accounts_in_message":[],"_revision_number":12},{"id":"aec22f825bb010480830a7f41d1c4c2b34d9fd9d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-16 16:42:29.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2903/.","accounts_in_message":[],"_revision_number":12},{"id":"cb7b1ef59cbcab224525744bb86b286f3b3e059b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 18:40:10.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fb85ea28f74721ab5eb9298d607e59\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4485d372f6924e5b89b3cf18c683c05f : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dddedd1d564c4677ada1e61d1c9f6526 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f20a22d85a384bf1b258c4d583c25b03 : SUCCESS in 11m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7def35ec0cb44599c5791f6f946a220 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b904b60fd814377acab941f508482bd : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/84682b8ebde44ae39bed406863d8e867 : SUCCESS in 57m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d5c2625b8654fe28ee74a4d0acea0b7 : SUCCESS in 30m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb05eb65a82741768f7ace725c024b5d : SUCCESS in 1h 37m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/feefe59632ca423bb1f3305b367ee8a0 : SUCCESS in 1h 00m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48f13f391ce34ed9a7085628d569117a : SUCCESS in 10m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/45317671cda34afe9f5aab9c899ad2b8 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d6ae2c16b3fc48b99432d5f0c4a050ba : SUCCESS in 2m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dcb6af332f8a49df90255dc372f83ced : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b310ecd56a52438cba558ebdd42545ee : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/166bf520f4a64bfbbc50fded6973974a : SUCCESS in 10m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0922355bd8fa436fac916d07031c6101 : FAILURE in 25m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c01c9dbd80954987a36aab154a6f89eb : SUCCESS in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4fac94dd4164bb1b7af160a332ba3d9 : SUCCESS in 1h 49m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/30e5149423d946d29ed49cbc3f98e7a4 : FAILURE in 3h 02m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e33bb889495f4eb9a15a2ff30d6b88cf : SUCCESS in 2h 17m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bbb2c81ce36c4be18d7e4962c4717173 : SUCCESS in 1h 53m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ba448c7deb584b9b88c61ff4a682f3b1 : SUCCESS in 36m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63c9122aebb74f2380d02f9c874a7537 : SUCCESS in 1h 07m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/580e2ab5b47f4fe38c7b43f27a1e741e : SUCCESS in 1h 46m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a4240ff070744738bfe612b8f7b6093d : SUCCESS in 41m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d381cb9c22784f858fcf5283466d60e0 : SUCCESS in 1h 00m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/fe33aed742cd439b891fd548c9a23cc5 : SUCCESS in 1h 24m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2403ef5e80894f8fb468ba9e04782fbf : SUCCESS in 1h 46m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad0df94532724edc8676bf1245465db0 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":12},{"id":"00f89471ba4bf28cbbe8606e4fcb3022205bb313","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-16 18:46:58.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_12_2025_18_40_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"f4f6b9f79e6425ef7e654e5080d07e31fb9ad271","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-16 21:36:48.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 36m 01s","accounts_in_message":[],"_revision_number":12},{"id":"cdf0e996c7d703f647aa3259ae753b3f84960970","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-16 21:45:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ec69df41c19f4d67be30cb550b15d0b3 : SUCCESS in 47m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/42f3aebfb4cc4c21a5e9f74713e6df41 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":12},{"id":"ce09b213a0607f3398ad6a62554c912eb4c7fbc7","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-17 00:06:54.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-933216-12-fc : SUCCESS in 1h 49s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-16-933216-12-iscsi : SUCCESS in 1h 9m 11s","accounts_in_message":[],"_revision_number":12},{"id":"a35df6dbfd717e4dd5b26b08fcaf36dbb0ddcac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-18 19:53:50.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"65e136e26552509b3d403c3c8b77202820b212a6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-18 20:07:24.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b06e3f54746c4daaa9fadaea91d99729 : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f28753cbfa004bd8a7b7e8b81a02af12 : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":13},{"id":"bd66c96028e09e5df85b03398b604c3954601076","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-18 20:22:22.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":13},{"id":"9be2f576ce213d10aa46622218aed2cf6081ce64","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:50:26.000000000","message":"Patch Set 13:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/750/ : ABORTED","accounts_in_message":[],"_revision_number":13},{"id":"68d41fbaa928e3b049fa1502a6b721f345044656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 22:34:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/342927376089489d865da94fdec2a31b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/142695f97c4b4b2e9d88dacbf7d2766c : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f76253298a64325b091247612e74563 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/240eb6c8f3b24ac1815f82a6ea43cad9 : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a5804dd62804270b24b557b2e60feca : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e37353b02a4810b2e22724b86d9558 : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/47c134e6ee6145d7a6c0b36e2e705e1a : SUCCESS in 1h 14m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6861aef2db34b9abe87c7087634677d : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96bc46041d074accb1f329c8912d4362 : SUCCESS in 1h 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18b8f69a69bc4451b485b67477f883cd : SUCCESS in 1h 02m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/743bc4b1190b42a786efe0e8eb20516d : SUCCESS in 16m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f328de4945e4d6d9619d6451aa6ec8d : SUCCESS in 23m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/33532b87659c4c9097fedc0139fcf2e5 : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e07e0cbfd4514c5fadc198f6e9c189ce : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/560107533c084b77b5597b5ae2afac69 : SUCCESS in 6m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3dbfb3ac1ad049d7bf408652938acd64 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a5c64dca3ef46dc95883c5766ba56f9 : FAILURE in 43m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6bd62d151ef246bfadf943eb408fcd23 : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c33fcaa2485d4fd293ee5fb0f9b81450 : SUCCESS in 2h 08m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/83f317df24ca46acbc180786897fd0dd : FAILURE in 38m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee2f83f381bf4434a93bfb024bc0efc6 : SUCCESS in 1h 51m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/455cc6853b7942ea8c86a0852fa64573 : SUCCESS in 1h 58m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7a997f6bf219408cbf3023f66e600589 : SUCCESS in 35m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ac694ff944c40deb581ead1a709357e : SUCCESS in 1h 00m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c5b55cb729a424abd17f3dfb156d49b : SUCCESS in 1h 42m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/429b9ddc8656423cb4841efd8314d48d : SUCCESS in 48m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f68a6791acd44757ae921b3b3fa61160 : SUCCESS in 1h 57m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/34528fdedf59440db51eadc479b875da : FAILURE in 2h 32m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b47c0fee14d04eb08920fa26b9c99a39 : SUCCESS in 1h 48m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67856393be00449396d1aed872c89e00 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":13},{"id":"0361b45edc06c68aa732c1c2cf3235b82a532ce9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-18 22:51:50.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_02_2026_22_44_45\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"f2551740c7754e8907f3804b311dbcd92d0365e2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-18 23:31:29.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/06ec77976feb4a3e974f88efff364eac : FAILURE in 34m 28s","accounts_in_message":[],"_revision_number":13},{"id":"2907d15f84ce6158b01c2a06a15a0457d21dbf9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 00:15:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5827213cc3c74a0dab3064e4cede8c73\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4ab1402c3634e0bb229b8eb39b1fcb4 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9a02a73aa274403bcf756ee8717221b : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ccec60ddd52d45b6a405ce74a713e617 : RETRY_LIMIT in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4bc4ad75f0357a9db236244f8efa76a309184fcf","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-19 01:24:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/e6ac4fb5bfca422d907a9f8962db4dc7 : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":13},{"id":"e4001cba70ccc8c5ca4566cf90a9f210e2bb1ce8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 05:11:35.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 38m 26s","accounts_in_message":[],"_revision_number":13},{"id":"c9560d1da84a0ffe1822e00774e4808c4f09b056","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 05:17:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6369ac9751b64b6e8dd36894fdfde034 : FAILURE in 35m 35s","accounts_in_message":[],"_revision_number":13},{"id":"feab17aaf44824cfcb8676e53e88cc6fd0cf49d8","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-02-19 08:53:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0cb954af480a8ed5da07baee9c7842ad48f5b242","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 09:32:04.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":13},{"id":"5fce2bc119e9e702d8e0ac318d90a02a954c082d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-20 14:11:57.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6282086122ec4c89f853607cd99896e18d07f1f3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 14:27:33.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4c8552e9ccff408884506c24d30bcc30 : RETRY_LIMIT in 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f711cb9fce7e437c938877889bab7768 : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":14},{"id":"94a977e985b3c729065f10208f4091bb6b4780e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 14:44:34.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e901aa78a0fd4159978014c74d486962\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d188d5ff1e3427ba683bd89792f6043 : POST_FAILURE in 14m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66187330f5404ae699cb297bb8a5fa8d : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc0cf49227f94eeaa164bb0eeca54b57 : RETRY_LIMIT in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"af4de76d1ce68e746b8edca589074270f2c7bdec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-20 14:51:13.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":14},{"id":"b1fd3ffe3549a3b9045b13510a90d9773bd22f26","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-20 16:13:17.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/eac9f41faad6440fb8b6dd356c9a1479 : FAILURE in 37m 54s","accounts_in_message":[],"_revision_number":14},{"id":"bc1257457eaac72ee41a83327399b2d08c1f129c","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-20 16:13:21.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver build/36ece3d833084a95a65a3b44366776f8 : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":14},{"id":"7846319b3b6ea964a42046bfd80acdefd8dd6efa","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-20 16:20:29.000000000","message":"Patch Set 14:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/933216 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"9ffb01e9b88b24ee4a6198ec729355f79915f420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:29:18.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7942e6b6a36846398d31300ce2aec6ba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a9a76510762e4360a3465b90c0cbecb2 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ebf71a7b364009b21d3f98b7537187 : SUCCESS in 7m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1388db0c13144c5eab90bbdd89442af6 : SUCCESS in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9338c592bd8241d6a9e27173f97d5b56 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25d87a03f76c4d818429329683330536 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12bd0d7e8884cb2b18356682f7c78ae : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/5b4778eb82364f7a8d804338d546f6b0 : SUCCESS in 1h 09m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f78c45834671496a91cd704083034058 : SUCCESS in 1h 11m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c2de476d54c4ad0b0b8deb240047d3b : SUCCESS in 1h 20m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eb0d80706294fb0b317d58f36f20ab1 : SUCCESS in 52m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dd5805018dd4919b2e824c80518ad77 : SUCCESS in 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04677e5e5d1848e9a30c600623c9d8e3 : SUCCESS in 22m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f0aca903b25149ea9a7bf975dee6e2a6 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f44bf851224d442983ee3424581adb11 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca654cfe1a1240c39558fdefd9af3ec5 : SUCCESS in 5m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/905aeeb43c7b4326873070819790a896 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d802055a660f4eeb9a412156c4065318 : FAILURE in 30m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4e4e72ebdf64591b65776fe7e5ce168 : FAILURE in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f850e0de7e8e4354b62997595f5795f6 : SUCCESS in 1h 05m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/10022fe3b50945a28c1937b192128395 : SUCCESS in 2h 05m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2305d7ba361541c396975ff506c93558 : SUCCESS in 1h 54m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/58c05d6cdf7e4ba0ab3cbd036c2a99fb : SUCCESS in 1h 48m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d594ce41673340f0824d8b1e3f0c10cb : SUCCESS in 30m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e82c20e4ada64762bdc6f7f105e47837 : SUCCESS in 37m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbfc669565d24e00890d15299b675e49 : SUCCESS in 1h 31m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abc5611c5a2c48a2b1b73c95b14c3e32 : SUCCESS in 41m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1573bac20c6407ab1c6037a7e237252 : SUCCESS in 2h 01m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e3830a66118248a99a8c445294f6fd94 : FAILURE in 2h 07m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbd700cb612c457d8d479740f0ae4017 : SUCCESS in 1h 35m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c226738270554d549e1a183b24398c72 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":14},{"id":"cc941ab1605381b45c0a290995c287d36e0de5c5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-20 16:35:39.000000000","message":"Patch Set 14:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/933216 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"8c2bbbaf1678ba4eb80bc559b7f0ac66c61486df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-20 16:51:28.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_02_2026_16_44_18\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"2ebdbd3bf406ebcc721b8b067fdb855544a0d97f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-20 16:52:46.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/e83814ff615643bc8445199f7531826e : FAILURE in 36m 32s","accounts_in_message":[],"_revision_number":14},{"id":"539ea45d152b6e1904a80ae1aa55e762d6945513","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-20 16:53:05.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933216 : FAILURE in 35m 16s","accounts_in_message":[],"_revision_number":14},{"id":"9cf30bbd96e018c4b36b6c4f75c1a317aefef8d8","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-22 00:39:08.000000000","message":"Patch Set 14:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_797/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":14},{"id":"cff4d26c365bd6fd70908d666332127cb230de2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-12 22:04:56.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"3cf440ec21b7302572ec47b5838d1a3c2d06a738","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-12 22:13:58.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/130b5aa55d834c769b9146a1601536ac\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7094d3474fff4a7ba7d67b060df84225 : RETRY_LIMIT in 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0461f8fcc0484d78861f759389cb06de : RETRY_LIMIT in 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97ae972410504aa79d633edb818b6e46 : RETRY_LIMIT in 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fccc01edd98c4767a94a3056c2970e43 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":15},{"id":"4f22b8cba4865a36490161735c32981b8b889efa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-12 23:07:22.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":15},{"id":"cec9bf87f800a9c2257e6757cc1519deaf94ff34","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-13 00:06:16.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver build/5f404e05b82948c09cbfda124a74c3ef : FAILURE in 28m 22s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/933216","accounts_in_message":[],"_revision_number":15},{"id":"d53ebf11a581f8c99dc0ee09b5c3ea384b6323ba","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-13 00:09:48.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/16ce606ea051440ca2e937c70155ed0d : FAILURE in 24m 02s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":15},{"id":"b80293bd181eced748f4d9caa1be7fce9cb26178","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-13 00:35:01.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d4aaac5a14b54cffa7daaf61fde22166 : FAILURE in 31m 36s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/933216\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/933216","accounts_in_message":[],"_revision_number":15},{"id":"c12e1417702102a9bdb504c69087f55f807b45b7","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-13 00:57:06.000000000","message":"Patch Set 15:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/163 : FAILURE in 2 hr 51 min\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/933216","accounts_in_message":[],"_revision_number":15},{"id":"218a1d34e69160e146b27701b053fc6ab6db05b5","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-13 00:57:10.000000000","message":"Patch Set 15:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/426 : FAILURE in 2 hr 51 min\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":15},{"id":"bc56fa0be4b4d5bc2d7be1b7ffa01e88689c9f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 01:35:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ee094b8da6142d3b826a2e6f8013637\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7c7c0bf78f804106a0ad56bcb7934c81 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1806c5563ca24d9db9bb751a84f27462 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4df17ff626e4dee905b70e422490e38 : FAILURE in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/800a7fee9a5d41ea99b1400f2dbeec17 : FAILURE in 9m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/63a82cb2ea914da3a4a6bd1c55bb2643 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df34e4c56b404407ae2b3a2d0b34f14e : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a1a16fe62e184b77a52e43296dc56f3a : SUCCESS in 1h 03m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/764b3d5ea53e468283af0400c346d1bc : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16d269a8d9814d6086b9d216f027a7a6 : SUCCESS in 1h 50m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bd15f93bf1541c2aaeca3d71dcb58bb : SUCCESS in 1h 05m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4540b3b223ba472f9ed4ad511cadaa65 : SUCCESS in 6m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a5a684753484c77bbba72a5666b7fca : TIMED_OUT in 41m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/59b53b5fb689435a9f8bf2ee04b92c96 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6406bcf60d7c45d59d5ee10c263d4521 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c279777212c54daba9d38d113810d1c7 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/05a3972c40784889af5afe138ebb890f : SUCCESS in 6m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/18b60404ce304d05a1de80b51619780e : SUCCESS in 9m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c019ca1271fd45a3a05f7458667699ec : SUCCESS in 47m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfe99622390944b5ad835a2c6028aec9 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0474fabcadca460d98771654c69f83b5 : SUCCESS in 1h 38m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c99f5150284449bdac7b2c74e34e844f : TIMED_OUT Host unreachable in 3h 19m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4424743f959042f384aae907e1a6e789 : SUCCESS in 1h 57m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0c2d17b0647d458c8acec73a5bb9448d : SUCCESS in 1h 56m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bb168b36caa24c019d46bb9f12039cb6 : SUCCESS in 39m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0f65f78cc1414275b47cd571b5e96290 : SUCCESS in 1h 16m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/feae29a3d538462495d3d2dd59eff012 : SUCCESS in 1h 33m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2db6b99d81704138af58837c0143f69c : SUCCESS in 45m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d337d70bd1b4602b9de5fbe2b53e397 : SUCCESS in 1h 40m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd5f343adeb64874940ceacaf95e8ce8 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":15},{"id":"4765bf37535125aa097fd3afdb62cba58afd1182","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-13 01:48:01.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7bb386579f514ad1b6ab576d6fb11baa : SUCCESS in 53m 51s\n- cinder-storpool-tempest-iscsi-multipath build/3632ed188f4b4c0e9b720851b285a4f3 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":15},{"id":"0f44351267e0c411028e5430f9be77d18ba6cff7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-13 02:40:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4470/.","accounts_in_message":[],"_revision_number":15},{"id":"f2867be1cb4c7d4642ac1bbf6952ce121ce345f5","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-14 13:38:13.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"23f8c4985772af3b394aecab866296bc616ed13c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 22:23:09.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb658b63d154784a0a6724436390c2a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52b1744178054d4f9607042c58232393 : POST_FAILURE in 18m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42a58d8ec3b84c3bbe8882c048b18f86 : POST_FAILURE in 16m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c5cb7867b874690b0bbc9585674ddbb : POST_FAILURE in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"32240c4a8686b24d21296c2f266760d1cb1393bb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-20 18:33:41.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/15/20260620183421/NEC-ISCSI : SUCCESS in 5h 28m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/933216/15/20260620183421/NEC-FC : SUCCESS in 8h 59m 17s","accounts_in_message":[],"_revision_number":15},{"id":"bee4fd2f2e1ef6ba5bbe32a861f485e6b1b1a0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 07:50:36.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"492da36c7dc756555c284d4a90d84f55d00304be","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 08:40:30.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":16},{"id":"d3f26ad55e70121bb09456aeb3c31aa296564fb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 09:21:41.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5426d6e211c542cdb8a7148f9c946ac7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d5f4374a08e44aab7a423088aed22e0 : FAILURE in 16m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ccae4cfef4db4d77944ef71b54aa0e6f : FAILURE in 11m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/20a3f7251ed2456eb3e7bd10500ba81c : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0fe385d1f7d92524c1b92fc83dd86bde72e85929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 10:54:35.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dc1bc19d294483e9cf3aad080ccf8c5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/72785f60b4fd4c85982e8f0e1b85271f : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ea70a327b34948a7a0f126545b0234 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9554b94786024789aee5edb8f6ee5751 : FAILURE in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/133028e8977e402f90ec19fcaed166a2 : FAILURE in 11m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81324e52ad4d44ada5d205006683dfc7 : FAILURE in 13m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921c19fba66f406b90394832fa27d7dd : SUCCESS in 15m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/426a9a5fd58c42d3910b84d14b3b97af : SUCCESS in 54m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b308de9a9c4436097d9382f4f00644b : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30aa8d350ad840cabb6dd3d6d1d67350 : SUCCESS in 1h 48m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d96704ccb64840b05c3fc928104b7b : SUCCESS in 39m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0417a72def54f009cf22d6348c244de : SUCCESS in 8m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17062689cb094ec192a30a295b0e66fe : FAILURE in 20m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c71ff115cffb4d3785e3e4e8668d98cf : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfc34faacd444cd59502655fa0b9c7df : SUCCESS in 6m 47s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/6d14f3f224d94b2db91fabe2f72c0b6c : FAILURE in 17m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20f2d77f5d63490e8b67093794bc0b98 : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b29f38012f0c4e3a815c39eec9e92d04 : SUCCESS in 6m 29s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1b63af623fb84fb79c077ec0e04c8d1c : SUCCESS in 6m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c4934aee67f4cfa9f846c0c271b112f : SUCCESS in 42m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b914cfc7e98410ea9b539bf0417619f : SUCCESS in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd244bde6b3c4223bc55d1c470bf1c65 : SUCCESS in 1h 25m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/653bc587e2cd4aaaa53b13e2cb98b539 : SUCCESS in 2h 18m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c800f13999264bdebc38d56571fbc3d5 : SUCCESS in 2h 02m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c4aff7c1454e441a9ab5e6061a4779df : FAILURE in 2h 58m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0e2a633ab00e49df85f55bb8456250f5 : SUCCESS in 1h 33m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ead2defbaf4d4a25a05dddee5364c1fb : SUCCESS in 36m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/56ca60beb3284108a6b08c31c8898691 : SUCCESS in 39m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e36c87d6965746e0a5835f4eaf9b50a9 : SUCCESS in 1h 38m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/db37fe0ddab14a2cae584751c6cdb3e4 : SUCCESS in 43m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9661a9e6885c463c90716bbcfe5aa149 : SUCCESS in 2h 03m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c31ff85753f347ec9aeb01722071f395 : SUCCESS in 1h 11m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"219717f962874f1d37909e8bff0322d2abe0746e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:53:17.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"d1c49edabd38cc60cffc376d43e0957f49b17176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 15:07:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb1b4cc29584c898d23269e829877e1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/732bb6397cef47b2a6e7bc4c195c688c : FAILURE in 12m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/93f57309eba5472393b35066ebf66555 : FAILURE in 20m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0240555a75a544fc92c605c8fd9dde02 : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7c164fa62eb04ce6543838096b7590a2200cafaf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 15:36:13.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":17},{"id":"0eed1e4313378d8a167c3f1b4d22fc044a8c492f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 17:21:22.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef5714be4597489ba790248416d876bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/534862a5e301412b917c4d46bc8516c3 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b76bf0878a4451aa90ae4b0432107d9 : FAILURE in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/802b150ee4d743f59e41f7a35fd4ad72 : FAILURE in 13m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0eebb5afa9649ceb8246139d48c1fd2 : FAILURE in 9m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fed2f5fbde6748329f5813b85a711218 : FAILURE in 13m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5386d554e564a7ba7eb7ab69c44403b : FAILURE in 6m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/8dd0f9f32bcd43a8955df67cc0ea4cc4 : SUCCESS in 57m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a8e68a1e85d47b78bf404a86f187504 : SUCCESS in 1h 07m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67f95c13f4a547fc873972798f8a9c8b : SUCCESS in 1h 36m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9bd1c11a2c84da6bae79e9358abddfe : SUCCESS in 1h 00m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b20aff096c4e4f77becb25336f603c62 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54516377b6574978baeb647f9425fe41 : TIMED_OUT in 42m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ccbd3a3cd9144bba56efb60b4847a32 : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebb64cccbb3e4413be189f1aa0bbe363 : SUCCESS in 6m 45s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f852003276d64df8847bb7e713118c55 : FAILURE in 21m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0cf73c00396143d5a9c235433eb66a82 : SUCCESS in 7m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8c4c64ed44654014828b393fa080f5a4 : SUCCESS in 6m 15s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5efd02e07f8b4aba996bbe7435c7e528 : SUCCESS in 9m 35s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/039624e9d51d4b5a8cda3f36c6d52494 : SUCCESS in 41m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d2a3b55ddaf41e5a556a038c192d6b8 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cba4397d6cd94870a1990d0167f0d012 : SUCCESS in 1h 45m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cec2907541a84a75b8db85a043451cca : SUCCESS in 2h 50m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7524bba0adbe47b3872211c443708e96 : SUCCESS in 2h 10m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e3767d0cb85d4145a0eed1a60e02676c : FAILURE in 3h 19m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/973733a0cb9242f5bfbe64e58cc98eeb : SUCCESS in 2h 00m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c317591b68d4c6d963a2646e0d664b9 : SUCCESS in 40m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ecd6ca5a5f549ff8267fae9fb3c6df0 : SUCCESS in 1h 23m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a6637f70379642bb9d0662d7556c6c27 : SUCCESS in 1h 44m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9c7eaa9570c641129da4f662dde4beed : SUCCESS in 32m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/425bfee88db246c09c14b6e1ec6726f8 : SUCCESS in 1h 40m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/215a8f28c6e34d40b900eb52835b2564 : SUCCESS in 1h 00m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file cinder/db/api.py","accounts_in_message":[],"_revision_number":17},{"id":"2cd12115cf4c7c2c109c6473295f7528a5a5e9f8","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 17:36:18.000000000","message":"Patch Set 16:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/392 : FAILURE in 0h 30m 20s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/933216","accounts_in_message":[],"_revision_number":16},{"id":"5f284fb1aea5ff7ffb2624fbfcebedad6e7622dc","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 17:52:55.000000000","message":"Patch Set 16:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/298 : FAILURE in 0h 31m 5s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/933216","accounts_in_message":[],"_revision_number":16},{"id":"894c8681a55bb88db36475764cd0022565751a88","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 18:38:31.000000000","message":"Patch Set 16:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/664 : FAILURE in 0h 30m 28s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":16},{"id":"83b3f6f08b3385101b39d38319f6b2e58720eeec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-30 22:27:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4588/.","accounts_in_message":[],"_revision_number":16},{"id":"70e362e964ff46a83a5e7133f8543ece79ed1921","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 01:35:47.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/311 : FAILURE in 0h 30m 8s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/933216","accounts_in_message":[],"_revision_number":17},{"id":"bd538fb50925d002b2295bddd238debadfadf539","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-01 02:29:34.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/be214bbd17e74e2fbbc9cf96761d814a : SUCCESS in 55m 01s\n- cinder-storpool-tempest-iscsi-multipath build/bada2cf2656a40eab220a5da9ecdbd5e : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":17},{"id":"56f1ffc9cd31fd2578a06eb1789c4e49e048bd25","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 03:06:09.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/405 : FAILURE in 0h 31m 1s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/933216","accounts_in_message":[],"_revision_number":17},{"id":"0b74801ceeadce021d51b324fc6b5353ef99fe26","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 04:27:34.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/677 : FAILURE in 0h 30m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":17},{"id":"86977e88f48683145d9887f6115d375034132897","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 13:11:22.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4601/.","accounts_in_message":[],"_revision_number":17},{"id":"25a77fec8535c1c9dca11dd494f0bc033d299a97","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 15:29:11.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4294/.","accounts_in_message":[],"_revision_number":15},{"id":"0d5ec728efc5dc04585c9a46f345f510303b0eea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 15:22:07.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4425/.","accounts_in_message":[],"_revision_number":17},{"id":"41c5f88b3396cde0edbaccbf0740209108148cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-04 07:09:15.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"879c874f04b64718126df93e6ef59fec6a8c1017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-04 07:19:01.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"c7236a4f73e0d2c4d3ed7678c931619e4b232a38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 07:39:06.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd1fb3a0fc49486b8378402855794f78\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dee3f9b3cba64f68a8280fbe2b857742 : FAILURE in 12m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/71ba782094d1467aaafaf3ae68047883 : FAILURE in 16m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cd1da82c753640e6b3f3958bc1ced854 : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"dc58959e98b3e19f0bdb6e3a531c0fbefe52dbe5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 08:00:58.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4836/.","accounts_in_message":[],"_revision_number":18},{"id":"52b1d9a2d940418caee491deddb24f7a7dc1137f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 08:10:34.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5012/.","accounts_in_message":[],"_revision_number":18},{"id":"ed34cd0cfa9ef0e0219d3251fc3f4b6ea240e926","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 08:11:39.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":18},{"id":"8e071a6bce457867f269ab364e7d96d4e7987d67","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 08:16:33.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":19},{"id":"e00a2ba9f49b6e66d16990ce4b3eb8d899b903aa","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-04 08:36:03.000000000","message":"Patch Set 19:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/710 : FAILURE in 0h 1m 55s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/933216","accounts_in_message":[],"_revision_number":19},{"id":"49b59c6653f807b2e013a30f62bd68bf71f1e0ae","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-04 08:37:01.000000000","message":"Patch Set 19:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1071 : FAILURE in 0h 1m 47s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":19},{"id":"a93e0ad35eb203561c2c21b82c82c5e79f7cdba7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 08:38:03.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5014/.","accounts_in_message":[],"_revision_number":19},{"id":"8d02466b54f47870a8f74655862bbf997b889adb","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-04 08:38:09.000000000","message":"Patch Set 19:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/803 : FAILURE in 0h 1m 35s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/933216","accounts_in_message":[],"_revision_number":19},{"id":"05519b8078e1c3699bcec83e82d1c8edc484a1a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 08:43:02.000000000","message":"Patch Set 19:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9c6d4a97fb444b12a2b9a3418e4a89c2/","accounts_in_message":[],"_revision_number":19},{"id":"410895aea025a8c5b12e9e0221fb3bcd207c0e71","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 09:36:35.000000000","message":"Patch Set 19:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9c6d4a97fb444b12a2b9a3418e4a89c2\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5f1a2984a1b44a889386945135b0db6 : SUCCESS in 44m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/947ce550631f425dbf26da19ae785db6 : SUCCESS in 40m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7e3200dd6dca4ab389c074e9430b4414 : SUCCESS in 47m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/638d2c1c71d94765b2faf4f64c423257 : SUCCESS in 36m 27s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f94c414788fb4128bf936961fdd83603 : FAILURE in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"52be2351d815c70040ceaf04b92176b32e5042cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:02:08.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/981712f7d69144d0959d631d25db4eab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d27d93908d4c4954b2345ab49783546d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e75c94f40e4483faa71a822b3fdab1c : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0000ba3b2134dc6b6f57a5b506c0dd8 : FAILURE in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/053d2d833d504e16b5d80363951cef2e : FAILURE in 11m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69acafb68d79413781831fd1e2716617 : FAILURE in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f390670dae8241b6a1fb49ac773138d7 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/58351003e9a84267ab497031c3562644 : SUCCESS in 1h 01m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a53b85260894fdaa594e1816b8406e0 : SUCCESS in 50m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/897dae99ce7c477f945ef7420d9b2f7a : SUCCESS in 1h 31m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c5bde625c284511b4641c44dacf47f9 : SUCCESS in 1h 04m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a84d3358c42c496f80d3f112a9f5a46f : SUCCESS in 9m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2258d823b12a45fd94ee4e57f867b910 : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/304565f6f4f94563a4da251bd46ec980 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dc9669b4854471aacdcea94411e05ea : SUCCESS in 3m 17s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/b3617ae93c814093b5f7eee7e2d95c65 : FAILURE in 20m 51s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c60930567d4e4c9f9e19ff0ad8565e32 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89f8f0b685ad428da339c90b69c539cb : SUCCESS in 5m 28s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0d34238346a94ebc864c1b59493f83ec : SUCCESS in 8m 31s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1e29975046f8440ab645acdcf45a647c : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17388048870645c49b343251b4a47a9b : SUCCESS in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f7eea8ce690c48cca2d902ea3f2c6995 : SUCCESS in 1h 35m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48e7e810ffb74b35888ccd1a6b675361 : FAILURE in 2h 28m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8c92a1ef8a8434f9e9978a919d65ff6 : SUCCESS in 1h 43m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/51993b7736664fea81b92b2b030e4642 : FAILURE in 2h 40m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4a48a60d37fb41a987a3affee641f14c : FAILURE in 1h 26m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3c8820556bb5429a99adaf1380c4ca19 : SUCCESS in 25m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ea88e534a714cd09125aa9814bfe27a : SUCCESS in 42m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3e25a85e0a3446e7b6ef95cae0d020ba : SUCCESS in 1h 32m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e2e21c4278e84f32a8d3c454f500e54a : SUCCESS in 51m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78f32bc0b5b74d0f9cc3c82134a6d163 : SUCCESS in 1h 37m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01ce4b461f07429b94906d1e3ebabae2 : SUCCESS in 39m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":19},{"id":"8fd7c810a3c1be9fe6b01a29bafc60e2edf290eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-04 11:10:04.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/32c0cc12dc36422e8b2c402de4917169 : SUCCESS in 44m 52s\n- cinder-storpool-tempest-iscsi-multipath build/c5edd8aec00345b8b8fcd982e7cd147e : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":19},{"id":"fa7d819c0cf61f2ccb5019ef59cc0d863bdf4613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-04 11:48:47.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"cc0e198d9f921d2bd06d597b6e450baba86934a3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-04 12:01:37.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1076 : FAILURE in 0h 1m 56s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/933216","accounts_in_message":[],"_revision_number":20},{"id":"497dc0ed1b2490804b532153fb34fccb3ef18880","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-04 12:01:57.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/808 : FAILURE in 0h 1m 59s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/933216","accounts_in_message":[],"_revision_number":20},{"id":"9369e5c4228c65a0fe88898e5db7725a6cb8fcf8","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-04 12:02:18.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/715 : FAILURE in 0h 2m 1s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/933216","accounts_in_message":[],"_revision_number":20},{"id":"595108132d3d163c6c412936dd34ab42b41735ca","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 12:18:16.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5019/.","accounts_in_message":[],"_revision_number":20},{"id":"bb1bc54895598c91699822c92ee34efc106b9afd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 12:21:20.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3af9dd7d2a4020bf7e953731185ac6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd5bda7f8a8343a59a280ad882678e2d : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48c4e515af094356aae792337fdf7fe1 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a6e43f392f64e1c8f0fb8401ce54c32 : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4c63bcaaa7c2899df1ea1fcd8ae71bdca28b6b9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 12:22:28.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4843/.","accounts_in_message":[],"_revision_number":20},{"id":"c0867b040ab599d3ae89dac39e7af6a0999628fe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-04 12:30:51.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest build/6777440add604f75ae1f640307c0a288 : RETRY_LIMIT in 18m 48s\n- cinder-storpool-tempest-iscsi-multipath build/2bb6a65bbf394ea48aa86188ca0a966e : RETRY_LIMIT in 21m 14s","accounts_in_message":[],"_revision_number":20},{"id":"df5e935a5ec4d60fc141010f6495715fb742464b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 12:32:49.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/69ca1b79280642e7a5ff3d6eff660356/","accounts_in_message":[],"_revision_number":20},{"id":"cd1ba0ef58dca1c86317e44c6e630f3363a6edd6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 12:48:49.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8 .","accounts_in_message":[],"_revision_number":20},{"id":"fec0ddbd72771175f6f7a4c1a3e7af3903c46126","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 13:31:01.000000000","message":"Patch Set 20:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/69ca1b79280642e7a5ff3d6eff660356\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65b49dd5b2db44af93449be2c7d1f1af : SUCCESS in 46m 39s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/47031e2b4ff6488497148fd0134cff85 : SUCCESS in 37m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ea7575b7961443095323e7c486098aa : SUCCESS in 55m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e98fa8086d0240038b4f1d01eb112523 : SUCCESS in 38m 40s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f5793707ae749f2a4d66617f3a615a3 : FAILURE in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8e742fab340c3e972b37c7aa6c9225c44bb14a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:28:50.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19df825af46d48eabc7923315f23d598\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3c18a44ea0b9448a96af61cdecae288e : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4237c6cbbba94c06a539e8a2b203c1e6 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/651e415da3fe4a939b7ad430df090673 : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56ac0eaba27142799850a383a9424137 : SUCCESS in 9m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5cbceeac09194c3ea85c9cdfa92be8c5 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2d3e85ee10495fa6e1df8e37ae8c9c : SUCCESS in 18m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/8dc349e66040425d8067cd5eddf4fc6f : SUCCESS in 39m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/985cbeb3cc20463698b6dae3bdf18415 : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0138a30657d040f7a2e523f5cb827d61 : SUCCESS in 1h 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a87c02d596e84505aafc88b6fee79e51 : SUCCESS in 54m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a5d4ae838364aeb85d4a4846abd8907 : SUCCESS in 9m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bf604b86516a41e5bd071eb98327cfac : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0b5b1953f76d42119634219c72259853 : SUCCESS in 6m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce8392b0c2514f9b91b8ab80880f9a15 : SUCCESS in 6m 23s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/19de7aa3cc9f498191254633710acc97 : FAILURE in 16m 00s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/82d0ba07ab6a405f89251814ee2e0acb : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5e392803ec3347fdbba8b38bce4d8567 : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4fc4e33f4b424cc59e409c7cb6775a65 : SUCCESS in 7m 22s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d154a74ea54d4b5b84e838adfe27cc88 : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12b72f185a1941449f6fd415f357322c : SUCCESS in 3m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/df4344ac93c14812a1f8dae8fd45b59c : SUCCESS in 1h 26m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3028e94859b43ed9e9c7330eab06a17 : FAILURE in 1h 44m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c892f745f65e4a7492be40db1fcf5abf : SUCCESS in 1h 20m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/923d84cb70834e7cbe5fd9f4e7a9abd9 : FAILURE in 2h 32m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a6ba5ffcf14e45d88d07ce4374453019 : FAILURE in 1h 50m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3b772fa755df4e2abcbd35dbde466d78 : SUCCESS in 29m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/634794288330459bbd8aa8cc6a4e9ed6 : SUCCESS in 1h 10m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9ccbbb5e09a4c73b5b18e8dd1e2a992 : SUCCESS in 1h 23m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0af031ab5763405ab9dfe70a1b4087eb : SUCCESS in 29m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1782478b54a648269a87284829136a29 : SUCCESS in 1h 26m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff28422b4f9431a98a5c31f1c4e9d53 : SUCCESS in 27m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"fe8f18112999a5971b157af9ae0acea2c8af195a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 14:45:30.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_08_2026_14_39_02\nLeave a comment with","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"8caae920f27c7c74dd36140d11cb57b931c21e3b","revisions":{"d23f4770dae0cf784b8e642cf756fac795318f4a":{"kind":"REWORK","_number":1,"created":"2024-10-24 09:01:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/1"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 09:00:44.000000000","tz":120},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at column for a backup so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23f4770dae0cf784b8e642cf756fac795318f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23f4770dae0cf784b8e642cf756fac795318f4a"}]},"branch":"refs/heads/master"},"f576c5c0d1544069c50d4c26b15efe44c902b77f":{"kind":"REWORK","_number":2,"created":"2025-04-16 12:59:49.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/2"}}},"commit":{"parents":[{"commit":"b74e8828e41885307618ea709cf7c51e3c2c83e4","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74e8828e41885307618ea709cf7c51e3c2c83e4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-16 12:58:50.000000000","tz":120},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f576c5c0d1544069c50d4c26b15efe44c902b77f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f576c5c0d1544069c50d4c26b15efe44c902b77f"}]},"branch":"refs/heads/master"},"aaa21e2499c4dd323d7ba747bf7a38b8462d6e00":{"kind":"REWORK","_number":3,"created":"2025-04-17 06:24:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/3"}}},"commit":{"parents":[{"commit":"b74e8828e41885307618ea709cf7c51e3c2c83e4","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74e8828e41885307618ea709cf7c51e3c2c83e4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-17 06:24:20.000000000","tz":120},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa21e2499c4dd323d7ba747bf7a38b8462d6e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa21e2499c4dd323d7ba747bf7a38b8462d6e00"}]},"branch":"refs/heads/master"},"be9e0af14af2fc7df99a9f5babe4afdd554ad843":{"kind":"REWORK","_number":4,"created":"2025-04-22 09:18:08.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/4"}}},"commit":{"parents":[{"commit":"b74e8828e41885307618ea709cf7c51e3c2c83e4","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74e8828e41885307618ea709cf7c51e3c2c83e4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-22 09:18:08.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be9e0af14af2fc7df99a9f5babe4afdd554ad843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be9e0af14af2fc7df99a9f5babe4afdd554ad843"}]},"branch":"refs/heads/master"},"7594c30640dc5a3e789df55b37c6c04bb7fccf37":{"kind":"REWORK","_number":5,"created":"2025-04-22 12:56:34.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/5"}}},"commit":{"parents":[{"commit":"b74e8828e41885307618ea709cf7c51e3c2c83e4","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74e8828e41885307618ea709cf7c51e3c2c83e4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-22 12:56:34.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7594c30640dc5a3e789df55b37c6c04bb7fccf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7594c30640dc5a3e789df55b37c6c04bb7fccf37"}]},"branch":"refs/heads/master"},"a1b4e4e0ad48a301054c59cfe42886a22d623507":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2025-08-29 07:09:45.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/6"}}},"commit":{"parents":[{"commit":"460f599abddc1d644a3c7284efdad8d8dadb5faa","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/460f599abddc1d644a3c7284efdad8d8dadb5faa"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-08-29 07:09:10.000000000","tz":120},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1b4e4e0ad48a301054c59cfe42886a22d623507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1b4e4e0ad48a301054c59cfe42886a22d623507"}]},"branch":"refs/heads/master"},"aeb0176c888623cccd79c532c65b38953f3b1d06":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-23 19:11:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/7"}}},"commit":{"parents":[{"commit":"49843092638aad6438d424c8ee73b78b6f09a822","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49843092638aad6438d424c8ee73b78b6f09a822"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-10-23 19:11:41.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aeb0176c888623cccd79c532c65b38953f3b1d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aeb0176c888623cccd79c532c65b38953f3b1d06"}]},"branch":"refs/heads/master","description":"Rebase"},"504b810762a780557f5f1ad2ba95cdd24bcc536c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 06:31:54.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/8"}}},"commit":{"parents":[{"commit":"f3f8a0c9cefc6ed5a74f4c21c632007ee6d8c2ec","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3f8a0c9cefc6ed5a74f4c21c632007ee6d8c2ec"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-10-24 06:31:54.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/504b810762a780557f5f1ad2ba95cdd24bcc536c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/504b810762a780557f5f1ad2ba95cdd24bcc536c"}]},"branch":"refs/heads/master","description":"Rebase"},"4ebc9f660c65f026150356d54621a7c3e311bd3d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-10-24 13:58:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/9"}}},"commit":{"parents":[{"commit":"e3639748c0ab8884a31eecaf8c62371223fad6d2","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3639748c0ab8884a31eecaf8c62371223fad6d2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-10-24 13:58:09.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ebc9f660c65f026150356d54621a7c3e311bd3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ebc9f660c65f026150356d54621a7c3e311bd3d"}]},"branch":"refs/heads/master","description":"Rebase"},"ce181433a92ac1118e8e9f82111b8daff60781b5":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-11-03 15:57:29.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/10"}}},"commit":{"parents":[{"commit":"5a7a441eca82f257ca013fcf7e017163ed308dbf","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a7a441eca82f257ca013fcf7e017163ed308dbf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-03 15:57:29.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce181433a92ac1118e8e9f82111b8daff60781b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce181433a92ac1118e8e9f82111b8daff60781b5"}]},"branch":"refs/heads/master","description":"Rebase"},"20ce98fb7c4274fa36c9f887c46fcaa977b8cb77":{"kind":"REWORK","_number":11,"created":"2025-11-04 10:42:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/11"}}},"commit":{"parents":[{"commit":"5a7a441eca82f257ca013fcf7e017163ed308dbf","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a7a441eca82f257ca013fcf7e017163ed308dbf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-11-04 10:42:52.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20ce98fb7c4274fa36c9f887c46fcaa977b8cb77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20ce98fb7c4274fa36c9f887c46fcaa977b8cb77"}]},"branch":"refs/heads/master"},"6cef4cb9d68ceb00e1be1f7bd1128e4a4e650812":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-12-16 15:32:30.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/12"}}},"commit":{"parents":[{"commit":"7016a22718773aa5239cde363bdabcaa87ea7ebb","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7016a22718773aa5239cde363bdabcaa87ea7ebb"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-12-16 15:32:30.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cef4cb9d68ceb00e1be1f7bd1128e4a4e650812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cef4cb9d68ceb00e1be1f7bd1128e4a4e650812"}]},"branch":"refs/heads/master","description":"Rebase"},"c3f7dc983b84ed896d9915d28dedcfffba96f308":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-02-18 19:53:50.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/13"}}},"commit":{"parents":[{"commit":"0e4175e53de71fcd258e2a8daa9d792187dffdda","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e4175e53de71fcd258e2a8daa9d792187dffdda"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-18 19:53:50.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3f7dc983b84ed896d9915d28dedcfffba96f308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3f7dc983b84ed896d9915d28dedcfffba96f308"}]},"branch":"refs/heads/master","description":"Rebase"},"04122e0727b011f4233de096d3912b8293c22c2e":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-02-20 14:11:57.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/14"}}},"commit":{"parents":[{"commit":"f3f5e7f02351c9add80c741cf4c66c79de4d80b6","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3f5e7f02351c9add80c741cf4c66c79de4d80b6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-20 14:11:57.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04122e0727b011f4233de096d3912b8293c22c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04122e0727b011f4233de096d3912b8293c22c2e"}]},"branch":"refs/heads/master","description":"Rebase"},"07883c22b1c8382496e463f161c53d4df0d27c15":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-06-12 22:04:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/15"}}},"commit":{"parents":[{"commit":"1b2b8c9ae564d2849f2a404b94b0852b56eb5c73","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b2b8c9ae564d2849f2a404b94b0852b56eb5c73"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-12 22:04:56.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07883c22b1c8382496e463f161c53d4df0d27c15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07883c22b1c8382496e463f161c53d4df0d27c15"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f3f5e7f02351c9add80c741cf4c66c79de4d80b6","ours":"04122e0727b011f4233de096d3912b8293c22c2e","theirs":"1b2b8c9ae564d2849f2a404b94b0852b56eb5c73","merge_strategy":"recursive","contains_conflicts":false}},"dbff1e41e304808667b8dd68d43f4a8827a9aab0":{"kind":"REWORK","_number":16,"created":"2026-06-30 07:50:36.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/16"}}},"commit":{"parents":[{"commit":"2cda620e03960b38a99459b215dd145a28cb8daf","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cda620e03960b38a99459b215dd145a28cb8daf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 07:49:10.000000000","tz":120},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbff1e41e304808667b8dd68d43f4a8827a9aab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbff1e41e304808667b8dd68d43f4a8827a9aab0"}]},"branch":"refs/heads/master"},"48cd018c215a8fd291b35572732ba22e280e1014":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-06-30 13:53:17.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/17"}}},"commit":{"parents":[{"commit":"1c7529c23705e4e515f38b7c532e27ddb75dfef0","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c7529c23705e4e515f38b7c532e27ddb75dfef0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:53:17.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48cd018c215a8fd291b35572732ba22e280e1014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48cd018c215a8fd291b35572732ba22e280e1014"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2cda620e03960b38a99459b215dd145a28cb8daf","ours":"dbff1e41e304808667b8dd68d43f4a8827a9aab0","theirs":"1c7529c23705e4e515f38b7c532e27ddb75dfef0","merge_strategy":"recursive","contains_conflicts":false}},"552c2e7da989e139e6cd944686106073bdc0cef5":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-04 07:09:15.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/18"}}},"commit":{"parents":[{"commit":"b44b101aa519044114e7d95111d6a0678d93cda0","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b44b101aa519044114e7d95111d6a0678d93cda0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-08-04 07:09:15.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/552c2e7da989e139e6cd944686106073bdc0cef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/552c2e7da989e139e6cd944686106073bdc0cef5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1c7529c23705e4e515f38b7c532e27ddb75dfef0","ours":"48cd018c215a8fd291b35572732ba22e280e1014","theirs":"b44b101aa519044114e7d95111d6a0678d93cda0","merge_strategy":"recursive","contains_conflicts":false}},"74c92bc8f4ef9d511af5f95ca9d9d44e3038d1d9":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-08-04 07:19:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/19"}}},"commit":{"parents":[{"commit":"576635b0e8109dab1425e61f7d2d9fc8d9748a50","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/576635b0e8109dab1425e61f7d2d9fc8d9748a50"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-08-04 07:19:01.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74c92bc8f4ef9d511af5f95ca9d9d44e3038d1d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74c92bc8f4ef9d511af5f95ca9d9d44e3038d1d9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b44b101aa519044114e7d95111d6a0678d93cda0","ours":"552c2e7da989e139e6cd944686106073bdc0cef5","theirs":"576635b0e8109dab1425e61f7d2d9fc8d9748a50","merge_strategy":"recursive","contains_conflicts":false}},"8caae920f27c7c74dd36140d11cb57b931c21e3b":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-08-04 11:48:47.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/933216/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/933216/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/933216/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/933216/20"}}},"commit":{"parents":[{"commit":"0098aea5cb422b5b703ca5aacc556650d9f06501","subject":"Add support for backup service to os-availability-zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0098aea5cb422b5b703ca5aacc556650d9f06501"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 08:39:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-08-04 11:48:47.000000000","tz":0},"subject":"Add finished_at column for backup","message":"Add finished_at column for backup\n\nThis adds a finished_at field for backups so\nthat we can calculation the duration of time that\na backup took by comparing created_at/data_timestamp\nand finished_at when a backup is available.\n\nThis is introduced in the new API microversion 3.75\n\nCloses-Bug: #2107469\nChange-Id: I6b3ba689c9e5dcdaa94e4cbf417103d578fe4be8\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8caae920f27c7c74dd36140d11cb57b931c21e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8caae920f27c7c74dd36140d11cb57b931c21e3b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"576635b0e8109dab1425e61f7d2d9fc8d9748a50","ours":"74c92bc8f4ef9d511af5f95ca9d9d44e3038d1d9","theirs":"0098aea5cb422b5b703ca5aacc556650d9f06501","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
