)]}'
{"id":"openstack%2Fcinder~933628","triplet_id":"openstack%2Fcinder~master~I60c61f184682e4a7e811fb10b9b69024a05bf16e","project":"openstack/cinder","branch":"master","topic":"bp/powerstore-metro-volume-support","attention_set":{},"removed_from_attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-08-18 11:32:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-01-09 09:44:13.000000000","reason":"removed on reply"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2026-02-25 21:19:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60c61f184682e4a7e811fb10b9b69024a05bf16e","subject":"Dell PowerStore: support Metro volumes","status":"MERGED","created":"2024-10-29 06:26:15.000000000","updated":"2026-02-26 00:25:09.000000000","submitted":"2026-02-25 21:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":125,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933628-bp/powerstore-metro-volume-support","meta_rev_id":"b3127bcd20b96738e9d168431eee50c023053fe7","_number":933628,"virtual_id_number":933628,"owner":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 21:19:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2026-02-23 15:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":2,"date":"2026-02-25 19:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":1,"date":"2026-02-25 19:26:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-02-23 15:24:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-29 06:26:24.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-29 07:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-29 07:48:10.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-29 08:36:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 08:45:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-10-29 08:45:52.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-10-29 08:52:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-10-29 20:20:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-29 23:04:18.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-10-30 02:09:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-10-30 20:21:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-05 07:18:55.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-11-05 07:37:29.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-11-06 10:36:35.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-11-06 11:00:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-11-06 11:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-11-06 17:05:22.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-03 18:16:57.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2025-03-07 08:35:15.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2025-04-22 08:42:08.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2025-04-25 05:44:40.000000000","updated_by":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"reviewer":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"state":"CC"},{"updated":"2025-07-16 18:21:04.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2025-08-10 12:38:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-08-18 08:41:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-18 09:36:12.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-18 11:54:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-25 10:02:59.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2025-08-25 10:19:58.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2025-08-26 10:59:03.000000000","updated_by":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"reviewer":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"state":"REVIEWER"},{"updated":"2025-08-27 17:58:33.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-08-31 09:18:22.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2025-09-02 05:53:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-08 08:53:55.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-08 20:06:43.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-23 15:24:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-25 19:26:55.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-26 00:25:05.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"}],"messages":[{"id":"559dfa8c025048e9270f2c579ba327bf0aa5b463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-10-29 06:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e94d72a02ca5aca8f837d1472247a4996957c7a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-29 06:26:24.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/820/","accounts_in_message":[],"_revision_number":1},{"id":"5f9e5de3bd8b078e080f3c27de838bfb682bc29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-10-29 06:28:32.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8b1c7654d16adf6078e9eb2b04fc41dd5ff12052","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-29 06:33:32.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"a208d0b89a6b539410f2163a64a7a37112b3bd81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-29 06:33:34.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/821/","accounts_in_message":[],"_revision_number":2},{"id":"a0ab62aaa9d5ea0d9e6772950b63392cd6593845","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-29 06:40:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"0f12a55866b2f98eeba9e6dbc9eb6578a71f8e29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 07:00:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/105798fe14ad412d8dc812ac65dad10b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc895bdcde444bf6bb8ab7d20dac284d : FAILURE in 25m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4443fc24ff74ad0a69b04889959b4b7 : FAILURE in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f677d5580171e1fa56d345e32e45fab9adc3a56","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-29 07:48:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/424a24e7f4af40ceb1e21610624d14b3 : SUCCESS in 1h 17m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bea86965cd9b44928b5f40a1b2fc45b5 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"544b9f932a56947cd376663197dcfec6269cdeec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 08:36:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf99079c7eb457086061bc26b4944cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7360b56a3e524334b3f03a40b4682a16 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f0667ee970438d938a703aad196932 : SUCCESS in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/568365ad2a524aad8096e98f517ee639 : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066ea0f6980d4446aa93976629a6747d : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/38db7eddd8074f4abd1842a53135c057 : SUCCESS in 48m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f5f7eb120a4d4ef9a3c1a7ba4a9914ec : SUCCESS in 53m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22e0ba3b8bd142369ed0f0fa5ce7b189 : SUCCESS in 1h 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e773e091da5547b299866c3479bb8a22 : SUCCESS in 41m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e592e6bb770b4902abf2edbd86f64bbe : SUCCESS in 12m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/27a5fad7447c4e0eb5b0d6103f58ecbd : SUCCESS in 19m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dff8e6f711a541a8a40fd3e89dd5d9f8 : SUCCESS in 3m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb0249cd449f408795ba057236084e43 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e610bf045674d48b24908366753ff1f : SUCCESS in 5m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7a7f736728ba49a796d2a66435cf3a7a : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8016e0bed9fe44ab8b6edfd1a4199768 : FAILURE in 16m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62ee7d7836914bedacb17c44c7c4c6cd : SUCCESS in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ba01618e38b74001818648277c5e1862 : SUCCESS in 1h 11m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9126905e373540be9eb1c672ec84fbff : SUCCESS in 2h 03m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78a8c6f2007a4c31b088cf53eb2452dc : SUCCESS in 1h 52m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/93f6b1278b104596ab69c8f8560a6416 : SUCCESS in 1h 17m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c33dec3fa1434d32803f0599c4b652b2 : SUCCESS in 24m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f60b703065a49f7a6f70542d7aed210 : SUCCESS in 1h 08m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca90bd9d7bdd4830bac4399bf0beca32 : SUCCESS in 1h 52m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/514c23048dcf4a0fbe497463b5d1cb83 : SUCCESS in 41m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/531fbe5647c94eb08f3827017efda3ed : SUCCESS in 59m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/23fbabc6eaeb4704b284b2dd7503eebf : SUCCESS in 1h 05m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab7f7d2f75e8486fa2c5c5dce6649578 : SUCCESS in 1h 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d3a4e05801840f8a4ce353a7f62c8be : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a226c2d475608aa279298981ed372bb4410f98a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-29 08:45:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c51ae8500ada49eeaa21e54b74e43058\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb382c329dcd41789f8428f940410cd1 : FAILURE in 46m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b60f6821c8f4c2c9dc8a2520736b99d : SUCCESS in 44m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f643f9b84b7542a6b395e6d8e7083a9e : DISK_FULL in 1h 31m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/187cfc8dd3f24dbea1b836227bfb200e : SUCCESS in 42m 16s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46853ca6f78445c3a4d3e734703ad050 : MERGE_CONFLICT in 16m 14s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26b343ff64754e38bcb56ce01212388e : MERGE_CONFLICT in 16m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f77854aeb85046e98cd2ca4cff569b6b : MERGE_CONFLICT in 16m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/755fb17c8f224790afda3065e3584eed : MERGE_CONFLICT in 16m 14s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} sourc... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"7cc06a83007234ae30534c80d25af6513faf950a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-29 08:45:52.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"6e59efdb85250129851b4e4581aef97656dc98b7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-29 08:52:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/f08741a2a1004908976480dbd5dc6d6d : RETRY_LIMIT in 4m 40s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/31f0a897c51e40828975dd13c6fb6bb1 : RETRY_LIMIT in 5m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"80529e213649342ab18f30647ad6922bad4ab780","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-10-29 13:23:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac05287371b25ef8df57952a9400687a8143adc6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-29 20:20:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/28/933628/2/check/DellEMC_PowerFlex_3.6/93d852c/DellEMC_PowerFlex_3.6/2314 : SUCCESS in 2h 29m 26s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/28/933628/2/check/DellEMC_PowerFlex_4.x/41bc7f5/DellEMC_PowerFlex_4.x/1194 : FAILURE in 1h 42m 47s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/28/933628/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/ae23121/DellEMC_PowerFlex_4.x_NVMeTCP/1106 : FAILURE in 5h 54m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"916e7985dd036e0e8aef495edaa1edf221a52101","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-29 23:04:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/2/20241030040556/NEC-V-FC : SUCCESS in 3h 50m 26s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/2/20241030040556/NEC-V-ISCSI : SUCCESS in 3h 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b09ac59b7d73f02744a9c27c08d0288604225506","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-30 02:09:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/933628/2/check/DellEMC_Unity_FC/3d2f361/DellEMC_Unity_FC/2807 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/933628/2/check/DellEMC_Unity_iSCSI/da65a08/DellEMC_Unity_iSCSI/2790 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a38376538e93fa9bd7fe40378a8cc83142657e42","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-30 20:21:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/2/20241030204922/NEC-ISCSI : SUCCESS in 5h 22m 06s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/2/20241030204922/NEC-FC : SUCCESS in 8h 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8312f50ca4fd16faa4aa483cfec186df7cdccc6a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-05 07:18:55.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c7c7a1be83483cdea170aea797301c024acd2013","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-05 07:37:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec2794f02922120ab0e4b6f2c0d29998d68a099c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-11-06 07:29:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bc1f55ecc4caa769637073586a2947abafcba0bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 07:29:46.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/867/","accounts_in_message":[],"_revision_number":3},{"id":"b15ac4b169385814771c265992ac3f4492969dfa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 07:37:11.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"f4fc7829e3be217282d573c1329dfe036f988295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 08:17:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d42cab9aac894cfdb4f7967caf6a5b64\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1aca3465bd024f70a8e4741aa45b9336 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/027451e85c114c5c94080302b49ef6fd : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c1b90bcfb365927e48a47922db2d04448ef7b24","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-11-06 08:32:05.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ecce12a3059af2909b2dc3ec53f980c9988b8f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-11-06 08:36:55.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f6f9ebebb9bef8ee2048da12dd61a934fddbf21a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 08:37:06.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/869/","accounts_in_message":[],"_revision_number":4},{"id":"b8610e6b4db433a0002b3b5f356ef53bcfe479ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 08:44:36.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"d6878f33bd12cd25ece94ac85217c8ed85562144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 09:05:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b486d81614ed48f18ea98150a612f87d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c5073438a0340259efb62442e4e9f9f : FAILURE in 25m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ca06b965ab04b0cbb637256330e2458 : FAILURE in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0811194aab2f3e8448499de6600869c5f8732540","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-06 10:03:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/24d969028bdc4d41a1891f6662d35358 : SUCCESS in 1h 16m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/576033f8ccc2490fa5f6a321e163d22a : RETRY_LIMIT in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"52de3070e3dc3c3c856e671fddf979c1b26fbdcc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-06 10:17:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/694f3bb604db4a99970ff6c9d1f39068\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0be556042e74ce2a1a41e021b299477 : SUCCESS in 39m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc647404530c4921a81b57bb64b93b8f : SUCCESS in 39m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef09eaceb7f4497fb733ed6c1352ac5e : FAILURE in 1h 40m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3e6883aacfc54bd186538cf36b1bbb10 : SUCCESS in 39m 11s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bb1392d98afc4956a3492d84fdb4475a : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da4b05b331794d56b9c6334e2cb8f5cd : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ca11fe27f624e83988a69cd0f8e4536 : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd53708871eb476faa64e4db76315f40 : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/d... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"3bd34baacb634de88ebd2fba48bdec8823d75607","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-06 10:36:35.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/822/.","accounts_in_message":[],"_revision_number":4},{"id":"d73724461d257fd6ddfbe19599b0765cfc5891dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 10:55:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c1bea388eb4f25b909c8f34d5ae505\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401a4a6249a44b829da19714e8afc63c : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed79f1a971184bccaca67ceb68fd87b5 : SUCCESS in 9m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ca92274fc0b4ca19695c10753f6402a : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0ebe5bc7c34dfb915004e416898a60 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5dec35e9741e429da275e10c5f9dcad5 : SUCCESS in 51m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b18f32e0433428e8c7f6039db719794 : SUCCESS in 48m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3125cccdf98946b28c0e1f76d1557326 : SUCCESS in 1h 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c20e4df4d78d4611819d646ccada6b57 : SUCCESS in 52m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45f71b69cdc04fc5932a998cd6a863d9 : SUCCESS in 15m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/565a803002504d1d88d9c468bc2bb612 : SUCCESS in 15m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aaebcab74eef435fb5951d9f4d433859 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cefecb5f6dfd44fdb23ab3cc45604220 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa331a796d2f4a6bac18e7e9efdc9557 : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/126bcf4584f94596bfe2cfcc7da50259 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d06040393e04be4b98c3588320aba59 : FAILURE in 19m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f9995eb2e494e5cba1db1fc5f7edccf : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ad0c970245d74d88b175b718ff9f294d : SUCCESS in 1h 42m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f58e308281ad4c6db014f888aaf4375e : SUCCESS in 2h 12m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d942a4f38495426a84ce4b9c5bc21215 : SUCCESS in 2h 13m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/af9311e8095245feb3795cb87d08796a : SUCCESS in 1h 42m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fd6a29eb88b8410ea2a97c663b136650 : SUCCESS in 26m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d90c9d2fb6c4c0ba89270957d5469e6 : SUCCESS in 1h 01m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad5ef6a4aa63489d8be83d5fd0e87ae1 : SUCCESS in 1h 26m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7cae89284a714f58ae4c990a4faa07ed : SUCCESS in 31m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b63cd91c78244bcaaeb4585223db569 : SUCCESS in 59m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b9d7275ea5b148c59ddb70817f55bf70 : SUCCESS in 1h 44m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7926dfdf923c48c1be25c0b6b475ad00 : SUCCESS in 1h 50m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d2e9e52b9047fdbb2cee8c4ed8448b : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ca4ab884e59337ceca70630c5d7a1986ba7cef74","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-11-06 11:00:08.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/06_11_2024_10_55_36\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"cab381ca3c6332aabcd584ed2af2e6d1bc9247f8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-06 11:00:10.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/933628/3/fc : SUCCESS in 1h 06m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/933628/3/iscsi : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f0a45274dffa25301984da97372287dd8c23030e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-06 11:02:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/d289c67a4a6349d4ae5f5d7919949b64 : RETRY_LIMIT in 5m 27s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/3a1f33cd49c44c8e9a71b63347ce0f7b : RETRY_LIMIT in 5m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"244ffad361a6fb37d3b31649de464aee6ab70d79","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-06 14:14:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1391/.","accounts_in_message":[],"_revision_number":4},{"id":"637f0d97ab57d56bcb2a7543a201697b8a2cb077","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-06 16:19:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/933628-4/Seagate-iSCSI/ : SUCCESS in 2:27:53\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/933628-4/Seagate-FC/ : SUCCESS in 2:18:47","accounts_in_message":[],"_revision_number":4},{"id":"24f5b45bd8c5de6036e0096e8ce46a5cb85a8958","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-11-06 17:05:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/d94ee5b23f504a529a3404a0371879bc\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/20a6baf4eb694ed2a2ed55bb764f5b13 : TIMED_OUT in 3h 03m 12s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/63598efe63e747f9b083f8aa0eb122f7 : TIMED_OUT in 3h 03m 21s","accounts_in_message":[],"_revision_number":4},{"id":"68f866b99006cf5f018dae0ef036828173189eea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-07 02:44:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/28/933628/4/check/DellEMC_PowerFlex_3.6/9ad3edf/DellEMC_PowerFlex_3.6/2374 : FAILURE in 2h 37m 36s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/28/933628/4/check/DellEMC_PowerFlex_4.x/a963df6/DellEMC_PowerFlex_4.x/1233 : SUCCESS in 2h 01m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/28/933628/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/1fc862f/DellEMC_PowerFlex_4.x_NVMeTCP/1135 : FAILURE in 5h 54m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"386ca006ca9b657f167409fb2584838d02993f6e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-07 03:11:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/933628/4/check/DellEMC_Unity_FC/937b582/DellEMC_Unity_FC/2841 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/933628/4/check/DellEMC_Unity_iSCSI/9e08aa1/DellEMC_Unity_iSCSI/2816 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"508abcfd80720e149dea04068e7d0a47c25e9af9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-07 03:36:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/4/20241107083830/NEC-V-FC : SUCCESS in 3h 57m 24s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/4/20241107083830/NEC-V-ISCSI : SUCCESS in 3h 58m 01s","accounts_in_message":[],"_revision_number":4},{"id":"185d36351f5cf8f17f092bf0969754d0d431fb3b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-08 00:35:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/4/20241108034714/NEC-ISCSI : SUCCESS in 5h 22m 31s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/4/20241108034714/NEC-FC : SUCCESS in 5h 48m 23s","accounts_in_message":[],"_revision_number":4},{"id":"b7d1deca666af528e1351c330cf0b717938dac76","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-03 18:16:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f937683b5510745f770e205b7c62ea5e7803d6d","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-03 18:28:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a385e057fb71f45938f499b9d85ad73fb26d180","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-03 18:38:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a4cadb33139287b3cbcc742716a71bc851b1ef38","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-06 21:27:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b7362934db1bf40c0c48f2f6dd64adf43e3e61d","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-06 21:36:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d51a8c70b264db6fbc111ef08871880995db42b8","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-06 22:50:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/b15fc63be36f4d4987819d4472eabc94\n\n- powerstore-iscsi-cinder-tempest http://localhost:9000/t/openstack/build/37fb20ac7f08421c9966cc0c690715bf : SUCCESS in 1h 14m 15s\n- powerstore-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/0259599789ae40c6a1081b33323a5d27 : SUCCESS in 1h 08m 32s\n- powerstore-nfs-cinder-tempest http://localhost:9000/t/openstack/build/19617bf1f98c4f7aa2ee54e6ba52a1a2 : RETRY_LIMIT in 2m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"1866f24b15a0b8d04d2de85ca46f46e6f5c6551f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-07 08:35:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"905d8b0ab004ba90b70f9cc29ae49f2ddcb0af43","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-07 16:52:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1dac0f4f7895dd76904e0c22dfd6905f4399a833","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-07 18:21:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://localhost:9000/t/openstack/buildset/e5739c661cd14453ab054af137b514b5\n\n- powerstore-iscsi-cinder-tempest http://localhost:9000/t/openstack/build/a5a74e2bc0304ba3913786d1cf69cedd : SUCCESS in 1h 26m 35s\n- powerstore-nvme-tcp-cinder-tempest http://localhost:9000/t/openstack/build/8c085be1b476490eba724637c49d4459 : SUCCESS in 1h 20m 49s\n- powerstore-nfs-cinder-tempest http://localhost:9000/t/openstack/build/2f26122c566b49f597cec076c8a6dd92 : FAILURE in 1h 28m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"2f2c2c22682370391d9131db4961092ef3c744ea","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 10:18:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d99b93a8809037e4bbd167773a101e40c020ca8a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 10:51:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/a947b45aa1344cfdb8f878215a2fa9c5\n\n- powerstore-nfs-cinder-tempest http://10.228.225.16:9000/t/openstack/build/bb428666e7ef4d8fae4e11d5651c0f78 : FAILURE in 33m 03s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.16:9000/t/openstack/build/d3a1078b0404473bb3da40b80070eb9c : FAILURE in 33m 25s\n- powerstore-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/1ce88e0fdce144de97ed99cce5425264 : FAILURE in 31m 42s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"eee742df70c731a0b10bb68cd91a65748ea9dc7d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-03-24 12:28:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.12:9000/t/openstack/buildset/9fd813871bff43e69d9041f4442e3c2d\n\n- powerstore-nfs-cinder-tempest http://10.228.225.12:9000/t/openstack/build/682a30fccc27443098fcd7246439e954 : FAILURE in 14m 53s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.12:9000/t/openstack/build/4513d504b5254daab8010c3403f73fcb : FAILURE in 14m 53s\n- powerstore-iscsi-cinder-tempest http://10.228.225.12:9000/t/openstack/build/7b4937816fcd4374ba458aeda015530f : FAILURE in 2h 10m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"e058fce4becd78f00e79b92df7404d66c962cf4f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-26 17:47:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9d6ac94b3dbf887778e87da521c95f0a2335772e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-03-26 19:44:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.18:9000/t/openstack/buildset/c5a3f7d6c79d4176aa03eb8418ac3ee0\n\n- powerstore-iscsi-cinder-tempest http://10.228.225.18:9000/t/openstack/build/dc3d868544d24f4ba3380cb48f5c6c38 : SUCCESS in 1h 25m 52s\n- powerstore-fc-cinder-tempest http://10.228.225.18:9000/t/openstack/build/d2e4eab9f7de48749ec6dfc4ac81a038 : SUCCESS in 17m 00s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.18:9000/t/openstack/build/22a1de921155448fb66dc79feaa7616e : SUCCESS in 1h 25m 46s\n- powerstore-nfs-cinder-tempest http://10.228.225.18:9000/t/openstack/build/86b8339aa2de49beb784355b2975ae11 : FAILURE in 1h 56m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"751285bfe81ba9ae1b9ff2e8559f02a20b90a315","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-14 15:06:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f350393c4ecd0f6cbaf12b9512de2dee0062250a","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-14 15:31:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2cac248296c0957f334fce8b046f8fab5c4b0e56","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-14 17:00:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttp://10.228.225.18:9000/t/openstack/buildset/10a0801ede944f5196f72ec8d24431a2\n\n- powerstore-iscsi-cinder-tempest http://10.228.225.18:9000/t/openstack/build/8a5513bd7649480bb4ae9e54aa409937 : SUCCESS in 1h 28m 52s\n- powerstore-fc-cinder-tempest http://10.228.225.18:9000/t/openstack/build/28312b0a25744125864e470b926c92fd : SUCCESS in 1h 17m 49s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.18:9000/t/openstack/build/e22d20071be446b68dec19e3164491d6 : SUCCESS in 1h 26m 24s\n- powerstore-nfs-cinder-tempest http://10.228.225.18:9000/t/openstack/build/0a28ce3cdcda4313b0e1cbf6a942ed15 : SUCCESS in 1h 18m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"d340adac6b38891361e57fadae639060f1f6b0ef","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-17 13:26:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"773228db0ca967ebda41421cd74f827a7121e33e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-17 15:17:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttp://10.228.225.18:9000/t/openstack/buildset/3d96249a244242fe9e00411350c98788\n\n- powerstore-iscsi-cinder-tempest http://10.228.225.18:9000/t/openstack/build/af5084ae23074ccb888a9781385759f4 : SUCCESS in 1h 51m 05s\n- powerstore-fc-cinder-tempest http://10.228.225.18:9000/t/openstack/build/d593293bf02b42a1808736dbb4038510 : SUCCESS in 1h 21m 01s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.18:9000/t/openstack/build/27b2325cbf3a4de79b460c9d19cf4412 : SUCCESS in 1h 33m 11s\n- powerstore-nfs-cinder-tempest http://10.228.225.18:9000/t/openstack/build/0d2c03e3e0614cd2bfe250ca0a9f6e53 : SUCCESS in 1h 21m 40s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"c8aa5c71af1f0e01c04d9122a5a635ceff552ea7","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-22 08:42:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"31dea12cff1f82f07ec244bb9e2b05e4da57c657","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-22 08:44:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a6f206880be3d016081688c07df4aa6d2b79e80","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-22 09:06:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"28e3917807c5ba72a904b1310c158a22bcc715c9","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-24 12:25:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab5a19c12440a93a1f1feb96b9edc84510baa53c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-04-24 12:34:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.18:9000/t/openstack/buildset/36728047296a4a5e9ef1ce4ed0b91caf\n\n- powerstore-iscsi-cinder-tempest http://10.228.225.18:9000/t/openstack/build/0ca31686dfec470bad4b63d296c7d4dc : FAILURE in 7m 11s\n- powerstore-fc-cinder-tempest http://10.228.225.18:9000/t/openstack/build/56472e272b5c4bae83a0e63f9ba3ccf7 : FAILURE in 8m 00s\n- powerstore-nvme-tcp-cinder-tempest http://10.228.225.18:9000/t/openstack/build/3c92d6d9793546178db7542232895a00 : FAILURE in 7m 01s\n- powerstore-nfs-cinder-tempest http://10.228.225.18:9000/t/openstack/build/6faedc6336d94c34b29b4ae84ff59fe7 : FAILURE in 7m 02s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"67bc0e041778966bde0025169fc91d9f4665fdd3","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-04-25 05:44:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd48c44e066f448120b4d0a20ee3a875470e946f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 05:45:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/4e14046760044b818e5b894ac6104e40\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/3c73337e671b4eca9e6f19d09aa5fa36 : NODE_FAILURE Node(set) request 200-0000113444 failed in 0s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/d8663f990afe4c7a8deff0d0d366e067 : NODE_FAILURE Node(set) request 200-0000113445 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3f2e454e7969cdcd8193e3c3564e948ef23ed16f","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-04-25 06:25:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d40f6324f232bd2b99d46be1a1f15c77b1bd28e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 06:26:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/c2aebbcad7bb4e42a15f1495412d1432\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/8f99fb02b1004a2f931ddf7302b46df2 : NODE_FAILURE Node(set) request 200-0000113446 failed in 0s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/45ef59e636524e50bd11b2e9ec978d65 : NODE_FAILURE Node(set) request 200-0000113447 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e1d37bfc77500cd9f929bbe715d01ccf008defab","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-04-25 09:52:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f7a4dc125a2b66315d5b9f7ed8119bd55553e28e","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-25 12:53:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4823aa9d37d98ec3bda95154482d12436f30d944","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 13:05:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/a4f5425b598244c7abb5f0d056d83724\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/d3c0e954c3d14001b8641bb00cb030a0 : RETRY_LIMIT Host unreachable in 3m 30s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/b1f3bd2c86064b7da977e96042011365 : RETRY_LIMIT Host unreachable in 3m 29s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b6cb0ba565954f8da9088498b3b8c9d9376c8638","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-25 13:08:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8fe231624c9d4789c7503766190b118f765f333","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 13:19:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/e538d4fb4fbe4584871e977668504885\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/b51d8239000544708cd810727b8112b4 : RETRY_LIMIT Host unreachable in 2m 33s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/91ed087f68a3441394ca79c2bd325dc3 : RETRY_LIMIT Host unreachable in 2m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"82edd05d6e442ba39bf02eebe78d41583548f49e","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-25 13:26:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da6c1092726dc7d03e31e58c2fc03bc078be405d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 13:34:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/565e52c33ed146dd814da950fe2183e5\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/16823b5df27e44c1a9ecfa588d1e3964 : RETRY_LIMIT in 2m 08s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/1763eaa1520741e8a335b2eb8ce56ff2 : RETRY_LIMIT in 2m 09s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"494b329ba0e08ceb4a104618028457a8b65022b3","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-04-25 13:36:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"84a0c5c31ddc9f9c3aecc3f8c2a50702d4e8f9fd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 15:49:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/2003fa6b098444cca33b100e343994d8\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/a51c4f6bcf1c41eb9be594c7f6865b6b : FAILURE in 2h 02m 03s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/9f2529bdda1f4f6d98ccb5e345c47fbd : RETRY_LIMIT in 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b4f03f911e870adc20e18a3e97f68523ef337d9c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 16:01:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8bec41172a3f8bbcfece0ff17f6a9aae5fcc8bc3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-25 16:03:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/2ba4636ee428444895e9f948218cce81\n\n- unity-iscsi-cinder-tempest http://10.228.225.16:9000/t/openstack/build/b5a9523aeb574c52b830e107d7d4d280 : RETRY_LIMIT in 19s\n- unity-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/530f6326d0e14b5fa7618aa6a1e5c56e : RETRY_LIMIT in 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"68e85dade84cae3344afde04c885f4f30cbdd323","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-28 12:27:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8e476529d9e1af5d761ada6b5ca4b22d918a981","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-28 17:11:27.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttp://10.228.225.16:9000/t/openstack/buildset/9bb53c4be73f4c4e8c13e39027d5d225\n\n- powerstore-fc-cinder-tempest http://10.228.225.16:9000/t/openstack/build/38ae6bd01ef642b88f7a55134dbf0953 : FAILURE in 4h 42m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"f1217f415849b8ca685bedd06ec9c20ea4fa6774","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-29 04:48:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ba20796f816d66f38dddf4c38f771b013aded321","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-29 06:00:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/b9870ecccc7d488e9d6a3e12e7191c5e\n\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/cd9ae1e0757148f79f9eff64aeb68f3f : FAILURE in 1h 10m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"238ebeffe933e6faf33b743b6ce30bc1c7c4dd9b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-29 06:59:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d40edbeaaac90dc0b1d76060c285ff1ae4f73c20","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-29 23:07:49.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/033fe84a0e154d03bff1ea83ed10b1d0\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9045c554333a48d48a9679341e7b28be : FAILURE in 16h 07m 42s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/bd8f03ac5be543cba785d237cb29e7b9 : FAILURE in 1h 03m 50s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"35c988b205fd56358220e1bbfbbcc1c06034effa","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-30 07:27:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ca64bced07d73c4928ed9fee4a1e607da24162b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-30 07:35:34.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/ebe05b0912c5445180386110a86fe751\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3381d2b6be644d768d4dafae219ef476 : RETRY_LIMIT in 2m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"59a114afa52d0c0b81f587e18dc237efabaca045","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-30 07:46:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c1883e3b7f9907aba15724b762a7858ffd9c106","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-30 10:38:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/56924ccca6b54e8dbe7ba0fc61c72f2f\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2a592807c11c4cdd9b57625a5ad42355 : FAILURE in 11h 30m 30s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a0c160a2c88841209a60f18a8dd2f42a : SUCCESS in 1h 13m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"89b9b0c7ae7638f4a8b0c08f8e2d8d1dcfc1f66a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-04-30 12:38:08.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/93d81a2d861043a3981a394c0193980c\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3e8aa73475054a059c3cb89e28d00d60 : FAILURE in 4h 50m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6fc0de8c03b5d5a7867740b0a0d96ae547b85aae","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-05 06:30:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d19bf9917963d75d9f1d8999bbe4582bd354afac","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-05 09:27:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/f8ff5f20c44b463f8f03c3e5f2f3f368\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/8ad88c9dd4c943a9b40fc45b527b5dfe : FAILURE in 2h 56m 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"91a25ebc4e0dfe8f2e114fa1ca2c0506a73e514b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-06 07:04:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6794417e335db60b7347aaafab989318fb36b33","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-06 12:57:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/a9632592d4a84ad8b63624f619341c8e\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/4d6f38a990f84cf9bfd8b74c493c6181 : FAILURE in 5h 52m 55s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"b4be58f7c873de7cd148e314cf28a598581e0f39","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-07 08:24:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bab3407c89939458519fa85a7701fa308430c94a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-07 09:12:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cabd7d0f6d55e37728f76927ef24ad9717e21ba5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-07 11:17:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6c6a614d91c74c189c091edc9bab9508\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/bff508b4ba4441bea85329914fb353fa : SUCCESS in 2h 04m 36s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf593c77d2250d729e029831511a14457b8aefc3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-07 11:34:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"180661e513b9f85bb690f497d94579ea7b26698c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-07 11:51:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/dbf76b73a0fd4ece8b5cd824f58b6f4f\n\n- powermax-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d4414010302d40e5bcc870687ae365a5 : FAILURE in 16m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"0012dd02ebc41cc2f759d70de2f382daee87480d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-09 05:08:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b25c252edc4235664cb8c2822137f704d753d28c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-09 06:26:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16ed59e28e116402d82bce255dc2e22e0f74c598","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-09 22:24:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/635b622dfab049909f9b106c73ea9a0d\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/238087a2a700431aa1c0fa6c49b10ab9 : FAILURE in 15h 57m 40s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"29446ff88716e8af948a9c0b563c3de086f44328","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-12 08:47:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"beb8873253d824c47d078c5e06f678a92150cf52","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-13 14:58:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11e0aa4676a29311adae521dbab986baf5a15a33","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 00:30:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/3cf01f079eaa4c999a3317407e13c03d\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/f2bf07de03c84af39ea1be002634a6a2 : FAILURE in 9h 31m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ed25035e9e456dfc64426553b1ac239d60fe4c37","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 08:09:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2c747e58ec54af2405e6310adb964e02622282f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 11:26:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af294c7a6ad071b41271b06a056cb6c040912ef9","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 13:05:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9547bd07f7d3c782aaf3cabe15a2259ba70f71a8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 13:29:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b3892789ff42e1f76bb5e46c046676ecb6741f9","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 14:25:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb4f935e4a29f167e744a9092fe49273e1cf18c8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-14 19:10:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/10d16ac960cc4355b7eaf04d44e5d584\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/472e80bb221d43869bdc4659b5447e33 : NODE_FAILURE Node(set) request 200-0000113618 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"47c66f5562bef54978f138454e6aa20cb9306ec2","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-05-15 05:51:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f23b76529bd69b519794f0ad1670c2f04c4fec1","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 05:53:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/0a786742486a4909b5a3246c1c8d5b51\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/499744b1ffd947778af1256c9af573f3 : NODE_FAILURE Node(set) request 200-0000115256 failed in 0s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/dcd1f667723f473297a2852ef97332b9 : NODE_FAILURE Node(set) request 200-0000115249 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c93ff4d4ef48025498dfa7a98c4f7fef6e7282e4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 07:32:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5eac93eeb96d27db925e0e9c81e3577946f8a13","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 07:40:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/69bc883caac14265b9afb003872f1cfb\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/10ef069082e24e448c86614d15779899 : RETRY_LIMIT in 1m 44s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/43fc4548d2ca40b9b364b7587970f715 : RETRY_LIMIT in 1m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"21de86e266bd1538706dad7e255ada5fd6ef3af3","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 07:50:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59abc8323541fa457aae02f76d33aaa18331bbd8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 08:24:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a5b319ed64d80978635a04302006e4b9d1781fa","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 09:31:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/854648a8ac2c4cf98632ab10e3e20117\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0a529c51d1064dea9f991ee937b26755 : SUCCESS in 1h 06m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"d0cfd59fcec3a64fdf30c4e4b8a725b77aff380d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 13:14:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"75267a76836be750bc5adbab0448fb48a420368c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-05-15 17:09:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39cf6f1a44b2abdd8f557530406f7db3061b0991","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 17:09:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8195bea0015194dc0d4e1539e4745ecbd7ec1ccd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-15 18:13:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/a959a18c8ef54b5784f5f88dc659eebf\n\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/31811c434fdd403589d892cb861655b3 : SUCCESS in 1h 03m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e1c9e7669ae2787e0b3ec5ff0672de7c304ee68","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-16 16:45:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d327a7df324253aa5a190ab61215d73bf1e87754","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-17 18:03:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/95354bb5c0944ff2acdfb1d2d8cb5131\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/cff5e1c2d66b46848a07a46f1a3351f1 : POST_FAILURE in 25h 17m 00s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/33efb70ef89b43ea9804d57bab74a73b : SUCCESS in 1h 57m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f339338734ea014c5638956fb5947435b4c3a020","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-19 16:46:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"71d221c7480a93ce51dfd00704165847e048d5f8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-19 18:22:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/a63679d3856d4cf8b42d378d10178276\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/27e780d685314368bfc1749956a22dde : NODE_FAILURE Node(set) request 200-0000115877 failed in 0s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/57b6612735924bc6ad5be2e23f28181e : NODE_FAILURE Node(set) request 200-0000115869 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"19ed21080e33217d48de1837d75ed2aef1fe05e0","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-20 15:09:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d59fe2d0d6e2f2a4cebc550ab459c012d8de8311","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-20 15:21:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/c1c3c1e8c10845e98d2ad448943f2c21\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/de876923cb3a40449c44bae925b0cda0 : FAILURE in 11m 49s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5368aafec19e44899058dd9910c094ec : NODE_FAILURE Node(set) request 200-0000115906 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"235b4acea0545339aae0a16e219085a7799fb449","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-20 15:24:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b5af225134540d664cce34194946ff8f1dec99e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-20 17:10:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6100f43167bd481fa1237159e7dc3d81\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/9155c392e374478c9d74f9886880b345 : SUCCESS in 1h 44m 49s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/37bbd0084f734e4f9c62b530554703f2 : NODE_FAILURE Node(set) request 200-0000115910 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"03df335be263ff27c599fe4f6abce96db55cee74","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 08:17:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c93e84b6a4c5640425aadf9a0708297a008a58c4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 09:52:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/1db48264a008488f8f21c27b724301c2\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/a99674337d6a42f29e007b70d2d2197d : SUCCESS in 1h 35m 20s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6adddce8554f49c6ae0937deeed3e75b : NODE_FAILURE Node(set) request 200-0000119290 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"0cc81806826d51554e66bde9c275014bd2e8e373","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 11:30:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e3859e921af9aa249c07f8224d6660e117b4b5e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 11:44:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/dabda71f65244d8fb65febeb399affc9\n\n- unity-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/3dc7ade843384fa3b2bc40f896303cb9 : NODE_FAILURE Node(set) request 200-0000119296 failed in 0s\n- unity-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d3d5357ea1ee4b5b9bc9553f161be9f9 : NODE_FAILURE Node(set) request 200-0000119292 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"30fedab6b7dfa10612008b5ea316210a46413934","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-05-22 11:46:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8582396ad8d64a2822dce99dcf3b9fa3bce70755","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-07 10:00:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"adf5eb3b1297de2fd89c834c52a13bd9c2f13e8a","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-07 17:02:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/c9dd027888c74bebb4867b5538e714d2\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/1874264f820c44c6b2dfec5c36768fa8 : FAILURE in 1h 25m 15s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/bc336d3078f44c26b5312f25afe2da1d : FAILURE in 7h 01m 57s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fdefd7802d694d8e8d00587ca5f0ce95 : SUCCESS in 1h 49m 13s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/e2bb413aa0264e8e8f8cd95c9e5baba2 : SUCCESS in 1h 23m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"5ae824b9009f0f7f41d8495c51613527e07d37cc","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-07-08 04:43:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bbfaa28ca4cd70f84983c5a1c0118f7da88cf514","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-11 05:17:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d6fa2c91f9aff0488c378172a6a15abdf2ca8bd","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-11 06:59:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/bd06c66c8f1346edb6bea81e25c9dab9\n\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/f98bbfa05ba24d3ba8385a54d9a36772 : SUCCESS in 1h 41m 04s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"7e71ef8a2c64e248b031a6a59df0317945facbd4","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-11 09:32:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c19cbfcac02a09816da7688259b21b5efceb9df","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-11 16:23:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/1337a778673c4da5a8768b2c5b2a14cd\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0021a00db5984072aac5f4824565df08 : FAILURE in 2h 04m 55s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/810bd3740f8642dca10f9ba53deceb8a : FAILURE in 6h 50m 48s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/0f9324e75bc6472c9b7c21c32b55497e : SUCCESS in 1h 21m 14s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/85e8d6bc792d454d85aefa5c140714cb : POST_FAILURE in 2h 50m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"f809c9bd24b9b27168112a4f37b2707cf2ecd72b","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-14 03:36:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"affb27d35611b4257696055b2ae076259b709613","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-14 08:44:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9cd1a38498555c2363a01c459dd15caa7324c0ad","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-14 10:22:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/2495dad5d942485186f4fabecb702ff0\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/8a7c3104aea74b89ab94113b87c7bfe2 : SUCCESS in 1h 37m 03s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/938b950447fe4eb698e60894e78fea9c : SUCCESS in 1h 11m 24s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/16b2acc180604e0cbd825eed5c38d682 : FAILURE in 1h 18m 13s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5067a77c2c644baaa48e606ec29c692c : SUCCESS in 1h 14m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"d3ee29752a2839b4162ab4f2cd529df17bda4824","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-14 16:33:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"659accd979c62c2a5b7e5ac25df268cba8b52924","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-14 17:06:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"edd858c608e28780d99b59e0b6d4cfddef3c3574","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-07-14 18:37:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6a44e6de93ae45cf8cacd77eadc20450\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/98e64860483f4a7dae5fb23fb80e20e6 : SUCCESS in 1h 30m 13s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d96709227e424d3388fbf8c1c5fb1ef2 : SUCCESS in 1h 22m 23s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/b637817c791b4823a563ce0208667acf : SUCCESS in 1h 24m 07s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/e3616261afa046b89d8e25892558be40 : SUCCESS in 1h 22m 46s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"79d79ee7442cd8cdbf12e3f38a4fcf7de97aed4e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-16 18:21:04.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56ae5c36b1ebd2069c50a8b153d8fada3efc00c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-08-10 12:38:18.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e1a6d3621277a3e7e9f41a5e0677958ecd98e127","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-11 05:51:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf06600180d3225599ac2cec4a7839e9fc9b35d3","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 06:45:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"83fe2ab4afa89d5d1abbf6b614b5127baceb3e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 06:47:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a37c9b7b23054434d710efc6d8d61f5f2af08234","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 06:48:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9e6558a7486d2e358344bc39f65758654c632739","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 06:51:34.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9ec4ff5a634836355f3cee91d7f9a6a868b84749","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 06:55:16.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/191/","accounts_in_message":[],"_revision_number":5},{"id":"0624fd5cda4f370835b474abbc7cf005eed5b191","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 07:08:27.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"6aeb0645a6c935deb8d08d3d80aaf16c40f45b6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 07:13:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2965238a6da54987ba6a7611c01bd7ee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98bd5eb84a07407abcbe7c542178916a : SUCCESS in 23m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95b022ecf2244c4384c0ead90c45ad6a : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c397741b16c342bf9ed9d2b4488c3ee5 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"878343bd866975ff48048f0e86b670d46e703531","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-18 08:24:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/fc072b25c4554275a1b7f0f1c371e62e\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5d421a712ec84a9683068a84da704f87 : SUCCESS in 1h 34m 45s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/5237cd5bc62042c08dfc5c3b1e1dadba : FAILURE in 1h 28m 27s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2384972102ba4bfba42e6da6229aeed0 : FAILURE in 1h 28m 44s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/2f40b61943fe4797b6e9e45554f8d942 : FAILURE in 1h 27m 05s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"7acc17f809437b0329d1c4f93f5fabd78d2a1f31","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-18 08:41:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/933628 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":5},{"id":"c0001168ad260935fbe26bcb3d4e0c1f4ef59dd0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 08:49:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1591/.","accounts_in_message":[],"_revision_number":5},{"id":"cf2246ffbaf1df85ffb11f574fc0e09df0c272bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-18 08:50:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62805ca502b94f58aab5c327d2254339 : SUCCESS in 46m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e35a847ebe39419dac4abca4f00988fe : FAILURE in 1h 05m 04s","accounts_in_message":[],"_revision_number":5},{"id":"627407b936538f7cdaededfc40e15d063461ae0b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 08:51:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1456/.","accounts_in_message":[],"_revision_number":5},{"id":"46637c9654a11eb01c262dd2e265fb2d1eb6fb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 09:12:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a13a4e6bbf456ea34cc5596c40af8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c32d80439f43abac9e6c0fe588630c : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2cc96c6e5b647c38e7c5c85b8695ea2 : SUCCESS in 7m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9628d5140d474670a5688f536e6ddd04 : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f0682fc1a6a4cc197c188e35877d445 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe49421e10d43089a6aa7d1ddd3a804 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/211d2b13fc5446cfa506a21f0682a047 : SUCCESS in 58m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32a58f15f097422c9ddbcb2c9a9e48a9 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8f33031a49a44faa52fd733c22ef2bf : SUCCESS in 1h 31m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a35f1688d35e49ee9423d06ab6edd53d : SUCCESS in 33m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a515d325ad848e896d637b58a5b3257 : SUCCESS in 17m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8dbd5d73de53401ca49352bb0fa50310 : SUCCESS in 20m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ea4ca6ebf3e449f89f1773bffed9e62 : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9a16cbc5b4e46d896be5ff4316874b3 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3d8b247babf4bea8c99804eda3f75ae : SUCCESS in 5m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e26b04d65754abeb43ee868b854a0c0 : SUCCESS in 10m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09eb0b0bd6854a458449561b6897b720 : FAILURE in 14m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43793dd26b1845e2aba9acb0a76a88c0 : SUCCESS in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59bec401f7804644a0614113cdf5c03a : SUCCESS in 2h 00m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e0484d9c43ab413cb6489f303de5a6d0 : FAILURE in 2h 23m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/118aca45c4bc402a8bfa1b8b537ffd15 : SUCCESS in 1h 47m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0345a8ee302741958f49dabfa193eb24 : FAILURE in 15m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a15b628bf2ea4e0dbaaf9e3103452499 : SUCCESS in 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb1b084119d84f9faf755bbcac628968 : SUCCESS in 36m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ebb814c516d14daea6b3de34e8f23c04 : SUCCESS in 1h 32m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/50dcdac3b2f44227852fdf471c84ec3a : SUCCESS in 30m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab3e58de651b4c1c8fb95842a036eb9a : SUCCESS in 1h 30m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9f9f2c4e18364879a2b3dfe5bfa09f33 : FAILURE in 16m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a64c3d7238f4f81af663fad3b6b6c00 : SUCCESS in 54m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b910a5a90b5240cd895bab296f9f86e4 : SUCCESS in 1h 34m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb5f060ae0d94132b1d47db7b4df3d6f : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":5},{"id":"aaffcf29d7fc0c61936df10ab8f805d70072d76a","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 09:35:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"454b52c8a226ef2ecb21fc2c40bc814e3ae28ae6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-18 09:36:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933628 : FAILURE in 1h 24m 28s","accounts_in_message":[],"_revision_number":5},{"id":"0c95015a33ca020f572a8bcd00ddc808c4d8c7df","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-18 11:11:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/6f4cee00ef7e43bb985ab6463d8a5c36\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/cb91e471fdd648aa9232a8450008d25c : SUCCESS in 1h 32m 02s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/27de2cf0ce764184a0b6454d80ce6c9c : SUCCESS in 1h 35m 44s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/09a39705e44942e78afaf1a13bbc33ab : FAILURE in 1h 34m 16s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/f484c3fbc42e4e3781c0d824801257d4 : SUCCESS in 1h 27m 56s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"a24e6040d9ad72c97e2081f5203d02ed440027f7","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 11:28:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95e77ded26c9a2d692bdd9a76b4577c3a4318d1d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-08-18 11:32:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"54394f4fba2f311bb30c10909f1d9520a7bb297e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-18 11:54:10.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-933628-5/FC/ : FAILURE in 5h 1m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-933628-5/iSCSI/ : UNSTABLE in 2h 10m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"c4dfb6b7f7a142b1276ae9ae87e4e977d7111b84","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-18 13:04:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/31f421903d8742648b9cb3aff52b9101\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/8754fd1f1ca848a58c23619ba2a93933 : SUCCESS in 1h 29m 58s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/adfacd2e268148069aa2e3e97f25f460 : SUCCESS in 1h 35m 44s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d236ee06661248ee82388009be77e4ff : FAILURE in 8m 21s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/bae889fa83f74a95bcffbc12c9581694 : FAILURE in 6m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"8ed972863cfda31291bcc2ac657e2eac08babad7","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-18 13:05:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6ab772e196b5d6d011c2e79abae1370eb354ff33","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-18 14:38:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/b5c42c6748844b1b9803c3db1994d537\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/19abb531935f4c1896748c8a3c1da0d3 : SUCCESS in 1h 32m 20s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/6c9f385e9efc4b9d9a2173828053571a : SUCCESS in 1h 19m 59s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d5b2d3702d354e49b07b1d0d93e98d15 : SUCCESS in 1h 17m 25s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/d0dc2b29533049c0b1094eb61bfe9050 : SUCCESS in 1h 28m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"dc713a9b7427d7e9daec3505c61ab54666839367","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-23 23:34:48.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/5/20250823222447/NEC-ISCSI : FAILURE in 9h 57m 30s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/5/20250823222447/NEC-FC : FAILURE in 10h 09m 58s","accounts_in_message":[],"_revision_number":5},{"id":"207a151d460c378aa1cbdbe31a20cb9c0d311df5","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-08-25 10:02:59.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"785ca9663bda831a5379576fefaba73f26a408ef","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-25 10:19:58.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"833f306202243bef4f8e875c2da9995455c61958","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-08-26 10:59:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"db95ee07a4cf622421e53c5c23de1dd6566d757d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-27 17:58:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4aa0353bff9a169914f7d3196062992ed82b8f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-02 04:05:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"85d8293428c478f9721be5f1b1127342dbb64914","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 04:06:00.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/389/","accounts_in_message":[],"_revision_number":6},{"id":"6f9aa4b629a9652e0ac310e2596a8709c0db60d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-02 04:07:19.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"5f825c0c4900cbffd26b93d5720fa8ffdae24b1e","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-02 04:11:17.000000000","message":"Patch Set 6:\n\n(12 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5e3f8b5a4605dcbbf05b46eaa250b1d67e578854","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 04:32:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d671869101544da9b90242ddcdd8ed92\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b564211d07e54318af29b8f1a4378e4b : SUCCESS in 19m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb72aa3fda81481a86f6d065004872bb : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fe85a431c6a47b58710e45a973cf51f : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f6c8f349d71191902c3008e48be2dce1be60d8b5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-02 04:57:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1f34c2ee9f0040e48e1b5025f08c2b79 : SUCCESS in 47m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6e8390214f8496eb9b1eb625b88e7fe : FAILURE in 50m 00s","accounts_in_message":[],"_revision_number":6},{"id":"4c60a769be642a94b38ce10a6b2dfd2ca22bf27e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 05:02:12.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/933628/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"9540316473c0cdfdba3c6a4041ba640a66c05418","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-02 05:05:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45da4994192df708f4f4b011589ddac19de3b452","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-02 05:07:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1776/.","accounts_in_message":[],"_revision_number":6},{"id":"00c1ba3579f149ee3bc007e1dc22cf7fdba8bf45","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-02 05:32:32.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933628 : FAILURE in 1h 25m 20s","accounts_in_message":[],"_revision_number":6},{"id":"7aaf6b17d4b80c99d3e660626a03f14cc94be7b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-02 05:53:07.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"6d526b3dfda3bee6ead28224069556b49a676ce2","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-09-02 06:36:39.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/b1c2d74010f84aaa9b33e2493d0bb5dd\n\n- powerstore-nfs-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/edafb119b74b4034b486e0cbadf2e8b3 : FAILURE in 1h 26m 40s\n- powerstore-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/82738ff585974466acfadc77944f952a : FAILURE in 1h 31m 05s\n- powerstore-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/ec24fe091dcd4d43a5fb1ee9aa7be3f5 : FAILURE in 1h 26m 39s\n- powerstore-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/7ef1670aa52b4be5bbf4965f3758ba65 : FAILURE in 1h 25m 02s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"15106f374bcf13d12835f8cad1914595c4f92a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 06:38:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2d820c7a534128b8ca93449ae19344\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1496d5ba706d4d6cbc3e51ff27a3fc02 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a1c35bd01614c0284d80820c6d1062c : SUCCESS in 12m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9290a5a51ef34d29bd7c6e3690b9c185 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de5081509189441094f2b281a98ea7d2 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f589ae6e9054819bdb915df32f9b4c5 : SUCCESS in 16m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/0d802d8e7e224a4bb6c0ce3ff8bec47d : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ae6e9b45d914fc99456752d18f59b40 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df99e74c6bb24d2f9bd79a48dcb46003 : SUCCESS in 1h 47m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/739de93402b04928adc4a94a79012c61 : SUCCESS in 43m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4feba85b4a54760bb1e3aab864ef2c6 : SUCCESS in 14m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a133795d91b24415b2d42fcf0f7bf00c : SUCCESS in 20m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ced57fc426b425db3322bac20d4aa4f : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a21a7dbe940404dad38434b01f244ea : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ffccb4501664910a7d074c338abd52c : SUCCESS in 5m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/458055aa9bf14dd394a8ca43c2dd6eea : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a69eeea2b34546479312ffa58dcd1d77 : FAILURE in 35m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/369c9da180b74c52ae4f7d6f87d559e5 : SUCCESS in 4m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a30224d99284fea8bd99995c22df3e2 : SUCCESS in 1h 58m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e5784f1cbc3041649e4918249e64b632 : FAILURE in 2h 23m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a0a242ae0eb44e5bfdfa98de5430d95 : SUCCESS in 1h 59m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5df25d4fe1ea4c98ba2bfad0c70db935 : FAILURE in 11m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a7dc542a499c4849b37ed66dc75a8439 : SUCCESS in 34m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91596997e1e142d680224621b34167dd : SUCCESS in 39m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2883346a3c30488e929a57677860d064 : SUCCESS in 1h 27m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f450b192a4cc4cb2be7509bb9b6f17a7 : SUCCESS in 37m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/019bd1f1c42d44f3906908f4c275e23b : SUCCESS in 1h 33m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b6059eb781b843afa4b892d375d34fa8 : FAILURE in 15m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2bdd2a40f3f41bdb93e1c481df81cde : SUCCESS in 1h 26m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a63874b6dd64b03ba70fb17cf595962 : SUCCESS in 1h 30m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/295e6705ecc24d1a88a3fa6ab06226fd : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":6},{"id":"a857d71ca2693e7a8f0c3c3f77435aba48829cb7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-02 16:47:41.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-933628-6/FC/ : FAILURE in 12h 40m 23s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-933628-6/iSCSI/ : UNSTABLE in 2h 43m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"f8f7797bbc953b0a283afce6bf3259ffbe1132d8","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-03 11:23:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c817d0b841a2132c65c7ddd178b746e29c9559ae","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-09-03 14:00:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0ba5c10a3d2c94c78b177162c49e9e61b9d14a18","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-07 15:07:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/6/20250907175049/NEC-ISCSI : SUCCESS in 5h 57m 29s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/28/933628/6/20250907175049/NEC-FC : SUCCESS in 6h 16m 10s","accounts_in_message":[],"_revision_number":6},{"id":"268c5c33d2280846d7ebe6945777e8cc1bee0dcf","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-11-17 08:42:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e6c81fd366c307bd72ddce6b135fc7e3beb7475f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-08 08:36:11.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e1a44a27e79a5ad22f9c6e401f30ea926694f602","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-08 08:53:55.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/143/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"48e677e81d2936bd71c16a4a15920a6c4b38c15e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 09:08:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c730a84a8144f7a5a50709ad8583cc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8886823c55924963a00482caf8ef748c : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4584f594f2ee4b52b57b72ae8f32d46a : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb207c40474b44f5ab239c43052df014 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3b23ce02c1b5621baa730fdefdc13fa2d41e8922","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-08 09:35:53.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I60c61f184682e4a7e811fb10b9b69024a05bf16e .","accounts_in_message":[],"_revision_number":7},{"id":"a2b98192feeb22e688b23fc2f7b2556b6385b04f","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-08 09:55:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/dee42f4a73d94930a9363f94bedf714e : FAILURE in 39m 52s","accounts_in_message":[],"_revision_number":7},{"id":"8c9edf0052090194c2bc938d3178036d76e8f301","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-08 10:45:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4e78f0b8b0449e5a9fd5c5a47166b56 : FAILURE in 2h 03m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/69d2af556d524bdfac5809bfba93daed : RETRY_LIMIT in 48m 56s","accounts_in_message":[],"_revision_number":7},{"id":"33a42d7a800e5364ea8215b19d6ae1d3e03c32a4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-08 11:17:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933628 : FAILURE in 1h 20m 48s","accounts_in_message":[],"_revision_number":7},{"id":"17b8edfba3396a56f999229da26ebf79b0eb4fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 11:25:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa2af37d3284bfcb43fd895414a0d2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247707bac4bc4876898e479a88dd46b7 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1446309b1675491d8f672528564bc402 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a814e5fac8f74eaaa2941d55d2214aef : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e781e4a2bb4a76870e73472d5f4100 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/68636fb7e2e44725904385a86f729bad : SUCCESS in 1h 09m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/476bf05d9e544f22993878aa13c1405b : SUCCESS in 52m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a016716e34f4e1ea39549e5763f9e85 : SUCCESS in 1h 20m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0d8506cf8194486928b17647573a8b7 : SUCCESS in 34m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d650daa3f345d18c30800ab6996778 : SUCCESS in 15m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b882d0268b4245e08ab7e9b83c6a619b : SUCCESS in 19m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b77dbd2775794264b5c30d854605d22d : SUCCESS in 3m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87ac0c68632c48da905752b5953bada7 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e278ddee4aa14d6eab7a5aa93ea8e3a5 : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/336dd4d8c84c44279c20e2b7694c50f5 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0e38c9b2eef48b4859f5966cd7d8a68 : FAILURE in 42m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9f97e7bfe274e41b06534cafffb1195 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4102cf32c9c941eb87226490d52188da : SUCCESS in 1h 28m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3edb02f5837f4290ac0f8034a24dbe7f : FAILURE in 2h 42m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e6342ee7f97f4e5882ed523a9ce9e89c : SUCCESS in 1h 59m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/76df3640d4754d00af2299344dbb4d20 : SUCCESS in 1h 33m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e6edaa1a95f40de8afaf91143710a3c : SUCCESS in 31m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/309207a531664505b75c6675800ac4f5 : SUCCESS in 1h 11m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/11211c7662774757b1de676a682ac4a1 : SUCCESS in 1h 29m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/21c3f54a662a406ea36f3b7752120b32 : SUCCESS in 47m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d81d4e8c2acc479094490690c7106eec : SUCCESS in 2h 03m 17s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a54332c8118b4a51a691f6bf0f6bdc49 : SUCCESS in 2h 11m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b1892eb7c6944e89e7d99caebb48848 : SUCCESS in 1h 40m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd3473af58874a0db94af8b79b5af88b : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":7},{"id":"1c3eaced8fcbafec05599849e39a996a36f966ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-08 11:36:51.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_01_2026_11_25_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"de83fac302fd7f22364e27eb11e6a2226800fb85","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-08 20:06:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-933628-7-fc : SUCCESS in 54m 54s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-933628-7-iscsi : SUCCESS in 1h 16m 2s","accounts_in_message":[],"_revision_number":7},{"id":"6e286e87c0e869e7acc54936df04017225f4b6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-09 06:37:46.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"caad3b7b5151bfc99bb2e8c2210b230773d540ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:23:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e714c75318848eaa9d3c3eb266377cc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e266814384d426b8dd307d890c1f02c : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/815e3d8c88574c8d832d802f37cd38a8 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b7b2de9f4a29477eb711288b02226193 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1a06823d7d299dcaca044b3a8d4d0d98a3ca75e4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-09 07:39:18.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I60c61f184682e4a7e811fb10b9b69024a05bf16e .","accounts_in_message":[],"_revision_number":8},{"id":"92c8e7557802348dd85126136b4185a34b9faddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:53:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1180aa30c555407eb472fe7578ea0ee9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bd1b007efd4e738c853261d48bd314 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b6f2009a174853bf4999c842bd4626 : SUCCESS in 7m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28c37b16d7dd4316b8d8ad57e1abe859 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e11e9f58b6433eb03fbebe17c1e882 : SUCCESS in 6m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/3456ee855dd44f7fabc69ed1f8a9cb30 : SUCCESS in 30m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c2471f5654c4a42b4bb9419c8a36d6e : SUCCESS in 29m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f93647345f6b496e88121c2340100bc6 : RETRY_LIMIT in 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7177284111074dc8805e396ed8febbe3 : RETRY_LIMIT in 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0786f14b3ff5425a995412d6162478cc : SUCCESS in 15m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c52156c790cd481086a792c4595081d3 : SUCCESS in 16m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b493a6bbe5240e4b9dc35ff8476c662 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7bd4ddbeba1942238230d2f4ec36e436 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62e5d23dd7cc4e80b84bf53519356ff2 : SUCCESS in 5m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eb784f6dc4024d8f95c2d10b0b2bc6c1 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4454fcdadeb1419dab776413c2fbeffe : RETRY_LIMIT in 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce9a43333c2e47a7827cf140ed919deb : SUCCESS in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db939312281a448e8c9918b7324b9700 : RETRY_LIMIT in 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d9857078b67741b091ffbfec26abb033 : RETRY_LIMIT in 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f4d2360e6cf04c318b81c3dd86c57858 : RETRY_LIMIT in 1m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/74a2f96fa3454501a9666dc9c2cb0dc6 : RETRY_LIMIT in 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2698a94435a842e6a7aceb1fc72911b8 : RETRY_LIMIT in 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1bc891a3e86642aa8fd481488ac62cc3 : SUCCESS in 1h 02m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45401f5a1cea412c8f6a5f78f7953b48 : RETRY_LIMIT in 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/16886b94d94341a4ac76aa98191d8978 : RETRY_LIMIT in 1m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/86ba5c7abf9b44f5998fec6294dbd485 : RETRY_LIMIT in 49s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1218553bd042471ebfc52f4c95a54f42 : RETRY_LIMIT in 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5da397d724f427da7a22a0fa50eba26 : RETRY_LIMIT in 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f42288b5a59743a894de8f491a9b253b : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":8},{"id":"fe04316207741d96ef71a898fc17f649128c6449","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-09 09:44:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c10b1ad677c34187b54d7b5e1494b073 : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":8},{"id":"91060a7fafcc032042fc296380dd72612cf85cb0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 10:10:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/53f9fcd6fd27400ba5137e368721c13a : FAILURE in 1h 08m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/939f5aa41b934f50ae12444dc111e030 : RETRY_LIMIT in 9m 36s","accounts_in_message":[],"_revision_number":8},{"id":"038dfe3678a3bb941214666ca690ed15625a746e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-09 10:48:59.000000000","message":"Patch Set 8:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/164/ : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"96b9b9f141f7e54fcc3967d3bec147a6127152ca","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-09 11:46:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/933628 : FAILURE in 1h 18m 38s","accounts_in_message":[],"_revision_number":8},{"id":"d020ba7e0b286213b9f397ba2dff89e368b0c4f5","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-09 18:21:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a19e0dd2a66517c63356cb971cd88206070b4e0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 20:00:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bdbbedfb5ef54e4488b3ff29b0d05aae : FAILURE in 1h 20m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2f75021ba0674988a49177f41a7414d0 : RETRY_LIMIT in 15m 28s","accounts_in_message":[],"_revision_number":8},{"id":"e775e08aa644997d15be2f2a3f68096f6dd9cf70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 20:59:50.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9ee12779b7b4a409cb138b2d774ef3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52af7acf4f541608b874613bddc1657 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7056645082394b00a70a56860ca63719 : SUCCESS in 9m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10433447999a4c63b9dc9947f2fbd3bf : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee803c8af2cf489a98c17dc8f6b712de : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b29803f3230f4d179f0cd3ac023b3d61 : SUCCESS in 29m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f998bdf2f4ff4b8a9e1918304b83585a : FAILURE in 51m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/067aa0ff2f8842afa7813f16239a4de8 : SUCCESS in 1h 34m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1bff52178c14e8e90a8b8092763052b : SUCCESS in 55m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3295405dbb04ed7bddd89ff9de36d25 : SUCCESS in 15m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c720161a40e4b9f866a7e8d733231a1 : TIMED_OUT in 41m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c97f28a7a06427a80855906f60963c0 : SUCCESS in 3m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5503a95fcb774130b3c9bb3ba5d5d987 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2932143fb32444c69d9abbf6a53e6f9a : SUCCESS in 5m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b2183ce6d994fbb80da6b69e765be50 : SUCCESS in 4m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe3b8ef63c074898a46fbf84c5e87e35 : FAILURE in 47m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cddeef54be7b4a4fbc17103bd2e1e6d1 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e55fd85f3a244f7fa0d7f190282c0e2c : SUCCESS in 2h 08m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1feec97df89b4a5c9022114a563a9120 : FAILURE in 2h 29m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af452867b2914e2da5da8e27abc68404 : SUCCESS in 1h 34m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/43ff6e891ce146f6875e40c66e895927 : SUCCESS in 2h 01m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/886f8a8f8d0640cabb9c164d1dfec495 : SUCCESS in 37m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6fd4877f5b6e4966acbd85a5b3d6067b : SUCCESS in 1h 11m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b2fcbcfe3c4941fd9f67d955550b9e87 : SUCCESS in 1h 33m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2610ae10115b400988e36a1a7cbfdcf6 : SUCCESS in 38m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c411836d76d4707b49fc098432700a2 : SUCCESS in 2h 03m 05s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/58bf9ebfe4204e1f823182e00ce2e43c : SUCCESS in 2h 14m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da576ef066ab43beb36d911c8ef87328 : SUCCESS in 1h 46m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4709c8e0e38843848476947d577bccf5 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":8},{"id":"e0246c21a480d7fbdd8f1f1f14cd8ba85feb5c10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 21:53:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877c2e468e5b49ef961a0d71c2986c92\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/917e5d4d613f4130b7a81ef263f8ec42 : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45d227b2665b47d88107ea04f7b8c0d6 : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad5cc1f45141407aae4b2da633b32529 : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a3ec8aaaefd5f12873f96698f29e1e8439a5edd2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-10 06:12:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-933628-8-fc : SUCCESS in 54m 40s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-28-933628-8-iscsi : SUCCESS in 1h 8m 33s","accounts_in_message":[],"_revision_number":8},{"id":"d5d4f8f120b4fe309951f3952605dda2c5b7d505","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-01-12 09:08:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6f872e3407a425097bc3125ec276ea36d2ad5b75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 09:40:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9842de01ee1e47c3ae5973aa957c2764\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0414a048c17f4d1ea71a6032e3fee6af : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/339f1631828c404f923b5e59407e320e : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0ffe0cc5fe746df8e44c259957e82ff : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"da18e83cd89af1a6986d6ec5e8409f1cb6447d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:23:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc416e67b4c436c9fbb54958559edf0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee58511522147ccbab90c4200f4d68a : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/118562e96a474f2181121f4947f52acd : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b7ceea9159940caa8a7c61f362ce77e : SUCCESS in 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0727d4e7a52f4141bc373aa9fe70e1ac : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/cb743fa2213844e6a44ffa18e77894d6 : SUCCESS in 1h 02m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb8056ff5ec346279d9091a6a7ac55ce : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c207c33b19344397aaad1d33b59ad97f : SUCCESS in 1h 27m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fddf7ef0404fb8b37a44e742b2a963 : SUCCESS in 56m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95e9a9d00b8640c5a2f608afbb0fd805 : SUCCESS in 12m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2823937562a4ac0be048bf68770b8c5 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9ff3aeb255984fe2b6fe1a749a9a9e84 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/509dc31545174099bd72d69b32f16f3c : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/519907a8e2514daaa00ade767373b49d : SUCCESS in 4m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/74d269b752ae41628e12e9f28e921aa5 : SUCCESS in 7m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0be185c6991842448065c127d0e373c1 : FAILURE in 39m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20392d047969424f8c0f1a7750c73eef : SUCCESS in 7m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f818636d0eb940a48675863761ec3fd5 : SUCCESS in 57m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/90ffa6ed20d545a5ae77d74f4b0057b6 : SUCCESS in 1h 35m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/126944044979469eb92831de846a0a94 : SUCCESS in 1h 56m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f0d438f6a17f483fb9b02f30d530399d : SUCCESS in 1h 53m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d5580a0b0d204eca8d7403fb2bcfd433 : SUCCESS in 28m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e58ad42164c4cda9e97550f54b354bc : SUCCESS in 1h 11m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6314ddf7243a4c469104ef4e7efebd63 : SUCCESS in 1h 19m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/445fd2998bbc4bd4b3d641eef771c270 : SUCCESS in 40m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d207b8cc62f240a08aa8a3fbd2965013 : SUCCESS in 2h 06m 24s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/09259287dda442f186476f831977b036 : SUCCESS in 2h 04m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59eeb4e174d0408ba8c783e953f6c887 : SUCCESS in 54m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5af14a72ad1042f5b3e187e5a143efa2 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":8},{"id":"029c0e9eec58b75180cb59096fb8404b5ba62ce1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-12 11:30:55.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_01_2026_11_24_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"88428048ee6149107b61ff195abca789a83b33e5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-14 14:05:26.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1425a65ba5824b42a5133c1a41215cb2 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d03d8c22967e49dbbdad9188a12af394 : RETRY_LIMIT in 15m 30s","accounts_in_message":[],"_revision_number":8},{"id":"10be5ebefbc1bea58c56232d5c19321cbda69f08","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-23 07:46:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cd99bd4e268ccafa431447b68b47632e8c51704d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-02-23 09:26:19.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/7be1b16529434da7ab899e66d3025faa\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/196f58486b084679bbeb4c420f5de539 : SUCCESS in 1h 11m 16s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/0046600cfc084d68ae5e77dd49f0552f : SUCCESS in 1h 18m 19s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/0008e294c6f54e13b532f4b46f77936c : NODE_FAILURE Node(set) request 200-0000054455 failed in 0s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8f7304ada14741dc8f6b96704580a63b : FAILURE in 1h 38m 09s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"7190e55cc1cdd62b62f410eb3f8d70241b020541","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-23 09:38:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f4bd9e9db03e5df339a87ddd344ffbcff3c2b444","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-02-23 10:57:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/01e0574de828480fa6eb0b223b7a4a6c\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/c0de3787fa7d47039fca0acf7851c208 : SUCCESS in 1h 15m 59s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9a0324f59b764d1e981d1503f9c3a738 : SUCCESS in 1h 19m 23s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/1a34c54507fa4d4bb42fe642f1882d93 : SUCCESS in 1h 18m 52s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/1f43407c00b04691b48beffa53e9d456 : SUCCESS in 1h 14m 34s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"77a315c1dfc8b61d5f8eed8cf9a3a9399c6da8f6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-23 15:24:56.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3ba747c95cd0989a9fab788514176a84db562d5c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-25 19:26:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"7e2ca014263651647704033798cc1f1bbd09add8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:27:29.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6a98b6e8b86ff55a4f3af9419ccc66095b7b7733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 21:19:31.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d85c6b200e24b5f8bf83d3bb9cc6cc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e16f0dfddb4129b2e4a1ba7ae12560 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c1a5417463644418a41f51aaa997a7b : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/288c4d95c3404eaf90d9ae95366cf8ff : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc5edf7e0a14d1ba9119dc11a55b344 : SUCCESS in 14m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c2edc96631cf4c4da9ff5c0e0a80e76b : SUCCESS in 42m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bc273724506474b9b9088a734ae70a6 : SUCCESS in 47m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eccf40c9871a413eb5f10032e0424fc1 : SUCCESS in 1h 38m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea6c15df425a4bacb748f178c232c328 : SUCCESS in 1h 07m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c9b5322b0d8470ca5989881734a49cc : SUCCESS in 12m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cdbe6b82b88943ffac65ac0354d210b9 : SUCCESS in 39m 01s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b875ed4d049a491cadc6b61e3619ddcf : SUCCESS in 1h 29m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/655d010927ca4d6ba190334f1633e915 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":8},{"id":"4c5ba6c5843592415311af3be548485f99445a09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 21:19:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"cc3b4c2458fac2152a2f6211a25a8513e0241120","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 21:22:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b703177c00d344729977594a62315474\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8898b4764b40ae811e7cb3b9cceee9 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e302e9e17e24d8bb0f1836ce1b96266 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":8},{"id":"0466633b17341eff6705eeb4e0363bd372059b9c","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-26 00:25:05.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_854/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":8},{"id":"b3127bcd20b96738e9d168431eee50c023053fe7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-26 00:25:09.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/854/ : SUCCESS","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8f2563d9f7bab5fc44d0abf70ac545b2073b616a","revisions":{"107326c36facc50b84e0e20245590f0458b1302f":{"kind":"REWORK","_number":1,"created":"2024-10-29 06:26:15.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/1"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-10-29 06:25:36.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes Added metro volume support in Dell PowerStore driver.","message":"Dell PowerStore: support Metro volumes\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/107326c36facc50b84e0e20245590f0458b1302f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/107326c36facc50b84e0e20245590f0458b1302f"}]},"branch":"refs/heads/master"},"1d4c37bbc851d18f7cc943a6c4167782e5ee2c52":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-29 06:28:32.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/2"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-10-29 06:28:32.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d4c37bbc851d18f7cc943a6c4167782e5ee2c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d4c37bbc851d18f7cc943a6c4167782e5ee2c52"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8703fc3dcfbcc379e23bf7065e286f9d07c682f8":{"kind":"REWORK","_number":3,"created":"2024-11-06 07:29:38.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/3"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-11-06 07:29:11.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes Added metro volume support in Dell PowerStore driver.","message":"Dell PowerStore: support Metro volumes\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8703fc3dcfbcc379e23bf7065e286f9d07c682f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8703fc3dcfbcc379e23bf7065e286f9d07c682f8"}]},"branch":"refs/heads/master"},"c63fb01e487613384155336f3664523b808b06f3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-11-06 08:36:55.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/4"}}},"commit":{"parents":[{"commit":"cd7a4da7afac2c9912404736399fd8e985be131e","subject":"Tests: Fix type error in volume encryption unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd7a4da7afac2c9912404736399fd8e985be131e"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-11-06 08:36:13.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c63fb01e487613384155336f3664523b808b06f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c63fb01e487613384155336f3664523b808b06f3"}]},"branch":"refs/heads/master"},"221e2ee4806041f2192c3574f371969dd39a3514":{"kind":"REWORK","_number":5,"created":"2025-08-18 06:47:59.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/5"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-08-18 06:46:45.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221e2ee4806041f2192c3574f371969dd39a3514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/221e2ee4806041f2192c3574f371969dd39a3514"}]},"branch":"refs/heads/master"},"e5dedfe58c44fd2ef93a57aef3bab4322ad12e37":{"kind":"REWORK","_number":6,"created":"2025-09-02 04:05:52.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/28/933628/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/6"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-02 04:04:48.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5dedfe58c44fd2ef93a57aef3bab4322ad12e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5dedfe58c44fd2ef93a57aef3bab4322ad12e37"}]},"branch":"refs/heads/master"},"68a58b41296f6bde0db0dfa06e3fdcb9b65b1e18":{"kind":"REWORK","_number":7,"created":"2026-01-08 08:36:11.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/28/933628/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/7"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-08 08:34:58.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68a58b41296f6bde0db0dfa06e3fdcb9b65b1e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68a58b41296f6bde0db0dfa06e3fdcb9b65b1e18"}]},"branch":"refs/heads/master"},"8f2563d9f7bab5fc44d0abf70ac545b2073b616a":{"kind":"REWORK","_number":8,"created":"2026-01-09 06:37:46.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/28/933628/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/933628/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/933628/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/933628/8"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-09-18 16:02:22.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-09 06:37:45.000000000","tz":0},"subject":"Dell PowerStore: support Metro volumes","message":"Dell PowerStore: support Metro volumes\n\nAdded metro volume support in Dell PowerStore driver.\n\nImplements: blueprint powerstore-metro-volume-support\nChange-Id: I60c61f184682e4a7e811fb10b9b69024a05bf16e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f2563d9f7bab5fc44d0abf70ac545b2073b616a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f2563d9f7bab5fc44d0abf70ac545b2073b616a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
