)]}'
{"id":"openstack%2Fcinder~935511","triplet_id":"openstack%2Fcinder~master~I762dc886ff2b774a45a2798f011ed2d9cfa9937f","project":"openstack/cinder","branch":"master","topic":"bug/2092066","attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-16 11:07:36.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_35676\u003e using the hovercard menu","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-11-04 15:27:17.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"37477":{"account":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"last_update":"2025-12-15 14:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-12-12 15:54:32.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2025-02-21 16:08:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_35676\u003e by clicking the attention icon","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2025-12-15 14:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}}},"hashtags":[],"change_id":"I762dc886ff2b774a45a2798f011ed2d9cfa9937f","subject":"Add different backup scheduling logic option","status":"NEW","created":"2024-11-18 13:05:31.000000000","updated":"2026-02-16 11:07:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"17465f895e49e3bfafef7c423adc2e620180fe73","_number":935511,"virtual_id_number":935511,"owner":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-12 18:30:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-12-15 14:07:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 13:05:40.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-11-18 13:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-18 14:04:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-11-18 14:06:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-11-18 15:15:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-11-18 15:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-18 15:32:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-18 16:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-11-18 19:53:20.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-11-18 20:11:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-19 04:43:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-11-19 13:18:11.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-19 17:12:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-11-20 03:25:04.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-11-20 16:55:57.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-18 14:02:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-18 16:20:21.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-01-13 19:51:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-01-17 14:18:19.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"state":"REVIEWER"},{"updated":"2025-02-24 10:38:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-06-12 07:25:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-12 09:09:35.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-04 15:27:17.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-04 15:27:17.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-14 17:06:34.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"25471ac9fc387e71519ba0e92c52ce8e5273978b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2024-11-18 13:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb04ffd07468ca1a7670e81d6e878bfada605a30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-18 13:05:40.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/940/","accounts_in_message":[],"_revision_number":1},{"id":"9f0ab4365387c6904d7d9d3c8d98188e0717c902","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-18 13:06:34.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"f0d07634d828ded01bdfcdde02647642479adf79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 13:23:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f9f9f573f74846847206c98e7657a4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3471970dab8041d48313a6f4415663e4 : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d886eeef800747f7b7e1be5d0e483415 : SUCCESS in 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f706a71456fcd4890dfebe92cc802729a28f764","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-18 14:04:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/880/.","accounts_in_message":[],"_revision_number":1},{"id":"97d12a3c9c1ddd98bd08f51b06ff667f5ed86355","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-18 14:06:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b8f81438c7da1352213ba141abeb2a7ce7b2dc93","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-18 14:41:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1446/.","accounts_in_message":[],"_revision_number":1},{"id":"cc88b2333559c40058f521a8db7cd494e715f5ff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-18 15:15:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6128843193414a8d8d3cb85603f00107\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53009963bc4049218bc0c1c440c4cc22 : SUCCESS in 37m 29s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac6765f2c076483d98f965bbc70452da : SUCCESS in 37m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a39ca7129dbb42f5b3eb786362d4c6dd : FAILURE in 2h 09m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d9c07f4985b4e068af045d69ff4d35d : SUCCESS in 40m 50s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc17339c6979427f8412b615abcbbd82 : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c88b807dbbeb41e0a797acafd35e7cb7 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6048d4fca7140799b5b1ce0c8cc31fe : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1367191de3824032b4bb32ee570c64c5 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"6e49bf101df3348820433e2a561ee047b5477949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 15:26:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4de281796da34f1c971e65d40c374b2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86daebe7ebed4442b9645d04a433896e : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1ed2bc4b1b44b7ab8753e3f08d30db : SUCCESS in 11m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9f4f83dc28a4dc697d3123545e2944a : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfbdd10c1212445c900248afc5f7c600 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/e62d9e9f5ce64a37b5de275ce96855fc : SUCCESS in 47m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82a1efdc0b8f4a5cb59e1fe590ef01a8 : SUCCESS in 40m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9c26b2bd7be4cfa94bdf50edb3afb22 : SUCCESS in 1h 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/015fe5ca00dd49f1810dbbb83f228c39 : SUCCESS in 48m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/81b6ddb3b34346b084561b462c5be950 : SUCCESS in 14m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/248282803fdb4bf6bd0a8c82675fde5a : FAILURE in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47d56b1bc5cc44b98eb7102515e56fb8 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab6bdd593f124cc8a378ea06d9b550f0 : SUCCESS in 4m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/807ef1a7e1944797b8d9ecca90a43f0b : SUCCESS in 4m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/123400789ccb492b9bcdbb0590a7b8e8 : FAILURE in 18m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bc4fd99a26c41a1ae657d98e66fec69 : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6ae6a9c496164f3fb01264913136fe79 : SUCCESS in 1h 31m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9920c1ad5114487a92f1ad38ef4b6026 : SUCCESS in 2h 19m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7277c70e03c4f3a9e9984480f090d56 : SUCCESS in 2h 05m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/50a98753243f4833b5f847fd03dfda0a : SUCCESS in 1h 48m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c29486648d34d628022ffb44c25adfb : SUCCESS in 24m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/674221c588684e5eac96129d2ff48650 : SUCCESS in 59m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/171eb9d074494fdd92e8b1122391e6c6 : SUCCESS in 1h 37m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bf71d959e41a4ebcb10fe93526981ab8 : SUCCESS in 50m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc8765664f584efabfd6b462061ee56e : SUCCESS in 1h 28m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d971bccb0c3d4e63a33927ad67765c78 : SUCCESS in 1h 02m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c35493afe2d453296a2ecb6c3da4238 : SUCCESS in 1h 24m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a71310846cd4242a9e5d3831a1750e1 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":1},{"id":"edbef04cbbe8bb9c9b4f59b9cb066969d6d3ac88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-18 15:32:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/afbfc934f3f24fc78bc57d49a54c5746 : SUCCESS in 1h 11m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d188bbea17614cf1a6a2f1a1bcbc234b : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2c4ec59600ba934f06aed1f9a90c45a7f4292586","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-18 16:00:11.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/1/fc : SUCCESS in 0h 43m 12s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/1/iscsi : SUCCESS in 0h 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"de6500158a3b63478b8e52da7cc319d66d30f789","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-18 19:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/1/20241119002540/NEC-V-FC : SUCCESS in 4h 00m 47s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/1/20241119002540/NEC-V-ISCSI : SUCCESS in 4h 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cae96f908f2761d7d7f92992813237db061c1d4f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-18 20:11:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/11/935511/1/check/DellEMC_PowerFlex_3.6/d87b4b0/DellEMC_PowerFlex_3.6/2446 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/11/935511/1/check/DellEMC_PowerFlex_4.x/960ae0b/DellEMC_PowerFlex_4.x/1294 : FAILURE in 1h 57m 31s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/11/935511/1/check/DellEMC_PowerFlex_4.x_NVMeTCP/e3b8df8/DellEMC_PowerFlex_4.x_NVMeTCP/1186 : FAILURE in 4h 42m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"13ca92951e6b118f09093e41e0169d15778c5d05","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-19 04:43:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/11/935511/1/check/DellEMC_Unity_FC/4644b85/DellEMC_Unity_FC/2898 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/11/935511/1/check/DellEMC_Unity_iSCSI/27a1184/DellEMC_Unity_iSCSI/2864 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"22982628f30da52aa19593e72f08e7f29176632e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2024-11-19 12:29:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"783c14b557ff661f49cdb25ce14dca29f0654612","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 12:29:27.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/945/","accounts_in_message":[],"_revision_number":2},{"id":"8ee32d25fc9704f45f8a79829ed330bf2394760e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 12:30:21.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"c019938202bec3e565d70efe7f1f07ecfeb286aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8547bcf326d4790a093230619ca8035\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed3e477971724aec8bd053d272174c84 : FAILURE in 28m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/171c4e7f851b4268a4387ac905e41bf7 : FAILURE in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2570edf9e55a7c66863790ddf417407326552258","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-19 13:18:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/1/20241119133237/NEC-ISCSI : SUCCESS in 5h 17m 03s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/1/20241119133237/NEC-FC : SUCCESS in 8h 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2ace88504cfe94cded9f514fe798e694c7162146","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-19 13:30:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8d38ecc454777d4c4abfb7fb39e65d0a2c466b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2024-11-19 13:47:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6925b35af8f9f027262197d930832450b7b4a276","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 13:47:38.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/946/","accounts_in_message":[],"_revision_number":3},{"id":"0bc4c666af13c368ba8ff753ccdb4b2aa4ff6914","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 13:48:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"30b715ba40858c030d8ea13b8d8014f25b2e3771","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-19 14:50:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":3},{"id":"0b9f84ec06f625497d2cee34ed4d5c0d3f478cba","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 14:51:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/887/.","accounts_in_message":[],"_revision_number":3},{"id":"b4761dd3f6faabccea2b509febfcce616c2f08df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-19 15:23:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc63081841024a59bd34dc798960c894 : FAILURE in 1h 12m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b48dfb3687ff4dcf8bc12fc624db2aa1 : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":3},{"id":"28e5085412726e35f6521ea7b97c7d3cfb9ac5fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-19 15:37:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3a99f1ae2fe34150944762a305efc3bd\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fed134c2a7da4415a56a9fa87a2b6daf : SUCCESS in 43m 55s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a827647358c4230a2f0deda895e68ff : SUCCESS in 46m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7a596614c1a482b8435ccbb46f95a1c : FAILURE in 1h 49m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25b6e2199b5e400d8b5ebcdbb7510e26 : SUCCESS in 34m 52s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc6c6426ae8e4f8fb85512f2339017c2 : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6330467596684cfc88be63872de56c5c : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90ba8328d8084dbeac565ef0bc7d6c50 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5a8ada972e644f1914d801b4b1381c5 : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"61f3d2072452c1c823c841b47e42228bea6dd9c8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-19 16:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/2/fc : SUCCESS in 0h 45m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/2/iscsi : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":2},{"id":"199ad698ef810734faec05cabc0e9cd3f8e8d89c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 16:13:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1452/.","accounts_in_message":[],"_revision_number":3},{"id":"693cf8c074f5c81efae722e223733a23a6aea5ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:51:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09c11cb5d0ff41068ccfff46015e4e98\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c7fd5ca37764b7da1eb0ab5fb39cc1f : FAILURE in 31m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52d7ccc45e944295aea5b82d9395c662 : FAILURE in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c535b916184d50c995085de599dab49e22bb4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:55:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a794c8c6a63841818fc0ac666924673a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab7ba671a6a491881b32b3e41e707ae : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b16194991db4f9d967d091b26dbb52a : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/511f0153a5b2403ba0373b2f3bbda228 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e033bffdcc542f99e57cde4127611da : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3b4bf517b2284aa289581e958c73d180 : SUCCESS in 48m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/afdb32741b494b95b63fc329e8ba533e : SUCCESS in 52m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0905a09b3a104a4f83e1a730ef2ef49d : SUCCESS in 1h 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ef11517180d43a4ac6af9fabc501bdc : SUCCESS in 1h 07m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78a35bf20ed14e1e9142b569f98b5e5e : SUCCESS in 19m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5daa057e61c044ccaf641fa569da5d94 : SUCCESS in 8m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12645a906e4747ce99ab645dd19401df : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dc89f7c43334e1882660a74b33251f1 : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/95e96d66b1d1436ca7f76c16f43b56e4 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/147ecf203f464527a89316a88ed0dc93 : FAILURE in 38m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1accf343ad142209dfb7a3d8aded414 : SUCCESS in 7m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/76a275696932407cbdf13730020db10e : SUCCESS in 1h 27m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99bc1eae500c41ca909458154316a337 : SUCCESS in 2h 19m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f22ffad281f4d3e8d01fdb3a427344d : SUCCESS in 1h 52m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0cc33474513245199cc8ab88e96ffacc : SUCCESS in 2h 39m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/098b8355deb7420493de49a2e74d1e9b : SUCCESS in 27m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00f602b5169f4b9daf0ef4aad153e9cf : SUCCESS in 1h 09m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0d92e28d3c5546a78d32321c3387dc78 : SUCCESS in 2h 01m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ad69e103e9244ad8a0fb43558d63fa2a : SUCCESS in 55m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c56b93c339854416bb4fc0a06a2a1455 : SUCCESS in 1h 10m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f392b73b29d440e2aadc4696a46026d3 : POST_FAILURE in 2h 41m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5047aea5a2542afbd5cd362fda8946a : SUCCESS in 1h 43m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee2069b4c544d78aacdf7dfff112fdd : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3b70826786ffa7b1b17d8b19a381f518ee68c957","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-19 17:12:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/744fdb080e574cbcb14ef19a33059ee9 : RETRY_LIMIT in 5m 05s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/8f33e3276f5b4df7a3630151e3c885ed : RETRY_LIMIT in 4m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"339adbb7c79718a4d010e89f6f79285a42fff9b9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-20 03:25:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/935511-3/Seagate-iSCSI/ : SUCCESS in 2:27:41\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/935511-3/Seagate-FC/ : SUCCESS in 2:19:53","accounts_in_message":[],"_revision_number":3},{"id":"1a1d99136f934084e54378ee34a864bc32277791","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-20 09:30:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/11/935511/3/check/DellEMC_PowerFlex_3.6/b314a32/DellEMC_PowerFlex_3.6/2454 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/11/935511/3/check/DellEMC_PowerFlex_4.x/dd4eb9b/DellEMC_PowerFlex_4.x/1301 : FAILURE in 1h 51m 06s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/11/935511/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/7c7ff93/DellEMC_PowerFlex_4.x_NVMeTCP/1193 : FAILURE in 4h 48m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"0ae9bf9cbe9ad2d7b3accefc18cf9fdda3e2de50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2024-11-20 09:30:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a714583d54d3eb0b605184c7a8a03d10b11facb3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-20 09:30:40.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/960/","accounts_in_message":[],"_revision_number":4},{"id":"f288bbef35c9700b80f9df2dba8cfc4bc467b800","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-20 09:31:33.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"c9e2ff40ea606da4e76dc4896123f6b31a0e00fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 09:51:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/913e001059d6420684f434266fec165f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f74090fd997402ca07aeb04c60b4194 : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46860fb70c9d422b9fd1c477dd8683f4 : SUCCESS in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"13cbd1416ce33d2137e3c87ceecae16c41052fc7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-20 10:32:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":4},{"id":"682fd3f6e7a15bf51d20c8fe2d5f12159da83efb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-20 10:41:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/895/.","accounts_in_message":[],"_revision_number":4},{"id":"5a3cf23ba2a38045ffb59cd02236b1aa28058005","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-20 11:39:52.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3815fd71519641f9927d15963a7e5a9c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55af3c76317548a893ddbbb15f1b2d4d : FAILURE in 1h 22m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75113c738952426db2c62334dfbebb3c : FAILURE in 41m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c20315605334ce8aa7caf39fbb7ca5f : FAILURE in 2h 08m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17a753f3cd504c4194d1fb032a4f5fdf : SUCCESS in 37m 15s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2845aa2e6a4c43b0a662b2321c73f9cd : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/358ec15f318a4a5badaff7ae3e43b760 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b8e468ab25fe486c874ebb4f426955cb : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b461b55bff0b4db3accc048f930b8ae2 : RETRY_LIMIT in 31s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"f970c45b3951e702cf75ad8d39b1059145cfce11","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-20 11:43:56.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1460/.","accounts_in_message":[],"_revision_number":4},{"id":"4cf851711708a995f6746a2d70e8fdfb5dda0e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:30:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51be780b0d534756a853faf6fbec3092\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe0e54fc29c47bb87fe2068cc2be1e3 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49ac86c23e2046b8bf91f2910f9ab81a : SUCCESS in 11m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4834b6c0648f4dbc878b82b10543b4f5 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d979223bd8c4f149fe67f8e6221798c : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ee871084f3db42258bab7bc4713e0a5d : SUCCESS in 59m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17094862d4de4b0bb2c70cfb9f0a4681 : SUCCESS in 1h 04m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0c468027d684bff9dba3a7aef42e8ed : SUCCESS in 1h 24m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a79733fe411249f8ad0a6e29c3b7241f : SUCCESS in 56m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08219deed26d470796b898e1443c1e50 : FAILURE in 15m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8f5e2d6f4211471c8e1f0e76726da423 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f09887ded014bc08b67aecd3eed1d62 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4cea023e02244b4bb7b0c4b0309f4d0 : SUCCESS in 5m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/078122558b0a48b39cc2d17e28eadc37 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/57321c6a65324d22938acfb6b6615cf4 : FAILURE in 19m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4f07a639ab6437a9d04c1f44e72ed97 : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49c1014b3f464783b9699cef79edcae7 : SUCCESS in 1h 34m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f448e730fdbc417e91933d3e9c6ca30a : SUCCESS in 2h 23m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee5c7e9c26cc4dc98535523d8cbc431c : SUCCESS in 1h 45m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fc91d67b0d8043ceb0f2fc0192a4bf4f : SUCCESS in 1h 59m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3edcdeca451845a493070f6b9eca7781 : SUCCESS in 25m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d910dae9adba4b948a98627beab870c2 : SUCCESS in 1h 02m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6512f8deec8f4d80922ce4042243f2a0 : SUCCESS in 1h 50m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7ef261789fe482b81d9373164db667e : SUCCESS in 44m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/edf49233584c4e69aadcb0a57587fdad : SUCCESS in 1h 27m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/61613a6f2d2e473784a5b6d852ff878a : SUCCESS in 1h 14m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d48422aeb8d246fabef720d7354dd729 : SUCCESS in 1h 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3eef03ee645417cb15a49d2c65f3e90 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":4},{"id":"510f896c8c0bde30b120e9ba46eea797d39860c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-20 12:33:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/79646fe0d3eb4e5ea857dcffd8b67970 : SUCCESS in 1h 16m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/58c67c75b6b34b59b3eea88be9d6ecc9 : RETRY_LIMIT in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"0f087c194b4d43ccf80270a20e4351d5f7d7f400","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-20 12:37:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/85e8450cf9cb44eeb45c611d2a8d3b9b : RETRY_LIMIT in 6m 11s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/20c44b86b3eb4f93938017d0df835c43 : RETRY_LIMIT in 5m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"75ddb01434796728619d5207eb98837fbc2ed42b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-11-20 16:55:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/0ab26622f2d641978f4874d92ce0247c\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/b07fe4af25cf41a5a552fb1353992df6 : POST_FAILURE in 3h 08m 08s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/88b403bdd65d4dd1b81797237b137a19 : POST_FAILURE in 3h 21m 41s","accounts_in_message":[],"_revision_number":4},{"id":"54187410be0513d102fe6998dc48476c52581e15","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-20 18:44:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/11/935511/4/check/DellEMC_Unity_FC/ba83e03/DellEMC_Unity_FC/None : EXCEPTION\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/11/935511/4/check/DellEMC_Unity_iSCSI/e750f9d/DellEMC_Unity_iSCSI/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d443758cf85d8a5e8e510455bacffc6728cbc998","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-21 06:40:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/935511-4/Seagate-iSCSI/ : SUCCESS in 2:24:14\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/935511-4/Seagate-FC/ : FAILURE in 2:27:38","accounts_in_message":[],"_revision_number":4},{"id":"8a74f506983328133f24382b7b4e70c9dcc2a4b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-21 09:50:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9d30a10ec93648f584d9dc5333244159 : FAILURE in 1h 16m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/30a3a55cc0e44d67ae3ce9f80ffab3fd : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":4},{"id":"064cf3e4d0c8e7d9b9be8ff426e07cdc5d9953c5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-21 19:33:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/11/935511/4/check/DellEMC_PowerFlex_3.6/5201b41/DellEMC_PowerFlex_3.6/2469 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/11/935511/4/check/DellEMC_PowerFlex_4.x/d168afb/DellEMC_PowerFlex_4.x/1310 : FAILURE in 1h 50m 38s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/11/935511/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/043b1a8/DellEMC_PowerFlex_4.x_NVMeTCP/1201 : FAILURE in 4h 43m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"bad197b3631fc7ddbdaa6729938e72e68c013221","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-22 06:20:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/4/20241122112029/NEC-V-FC : SUCCESS in 3h 58m 52s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/4/20241122112029/NEC-V-ISCSI : SUCCESS in 3h 59m 35s","accounts_in_message":[],"_revision_number":4},{"id":"352dea0b5b2b47b8a44ae6dd4c5df5b335c20cae","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-22 16:04:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/4/20241122191401/NEC-ISCSI : SUCCESS in 5h 25m 00s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/4/20241122191401/NEC-FC : SUCCESS in 5h 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e2104822c8362a3cd9310400b2158b480e8660a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2024-12-18 13:18:12.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0baa575808b7e66a384633f2f8e8df8d1d7c0ff8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-18 13:18:19.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1116/","accounts_in_message":[],"_revision_number":5},{"id":"83bb084c2100e2824bf5e13f28ed3028ced875a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-18 13:19:13.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"7a694999a784dca003b03d27fa0a8fede2ec98ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-18 14:02:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/935511 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":5},{"id":"dd3d1417463a9ba6236329961980444f2a05078e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 14:06:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45763e1fc1f44653b5252a98def6c0a2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee4e64377cd54ad6b1623868cac35d89 : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb3b298920304fdb9e21267576cd9d0e : SUCCESS in 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e95264704722923da0b7ea84860e6079ba29ad8f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-18 14:20:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/144/.","accounts_in_message":[],"_revision_number":5},{"id":"edbb669045e2684f1a0796b4431a3b368cacb283","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-18 14:38:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : FAILURE in 1h 18m 35s","accounts_in_message":[],"_revision_number":5},{"id":"ef7aa0099407e660bac802cdd7d29cae50085a33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-18 14:59:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8f62c63a215e4481892fdea24f6eb770\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a218ead5f9b247878933205bc8f2cdc5 : SUCCESS in 41m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5c705a7da884e6e83b4edcf656774dd : FAILURE in 1h 01m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ec5e89a8d2f498e8835cb367b9a4d87 : FAILURE in 1h 40m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/48e69022c79742d0a4a49e97ddf6409e : SUCCESS in 35m 56s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb5daa7f00414ec2b6d9feb6451d90ed : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40698083753645fca36c97f8a016e8d2 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/010fb5407e944eaa9bc39473b61065db : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cd678e780a74bf1a028e444534d2354 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devs... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"4fcc0f0a3fc869fb6d1e8c097e76727d1848f269","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-18 15:01:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/134/.","accounts_in_message":[],"_revision_number":5},{"id":"6c7fc7d6914a24b3420c80488a24c050399631dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-18 15:52:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cc892c0ad2f445bd8d84e7c159f143c0 : FAILURE in 1h 21m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/af843a3afa92404487788b8846645eb2 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":5},{"id":"3e7f371e4f38a4a5bb36f367a6b92433cd891300","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-18 16:20:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/935511 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":5},{"id":"85437bbe56d87aacaf0c103be345d1d8bc30cc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 16:42:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c5f68ec3ba14266b00721e22a930277\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68ab51dc4da467f82c0210839a05c57 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73c44b7a6e5640ffbd7aeaa22d12712b : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cfb180ffea344b1a77fa64d7805e28f : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3e55a436d34b26b1faa5653712e943 : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/1a8901d27b724cbfab87a619557529c4 : SUCCESS in 58m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ecf922fc63c429eb16ae03baa263a2c : SUCCESS in 55m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8d166b98122445d888768a51e7d9c8a : SUCCESS in 1h 24m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7773b0e142e4f869feb740ac126bece : SUCCESS in 1h 02m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/750c06571f0b45dea26b700e33f73c99 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2114bfa8f43443c4bb451196b12922d2 : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f33697b72ab496192cecdfe58a0b30e : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/edb65548839147dfa7ed2c8c50c01308 : SUCCESS in 26m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1da78f9224cd41b3b5223b8c7e85a4c3 : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0a0d513c08b4b3dae53284aabf858de : FAILURE in 22m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c95e3e068c364cef90d826fdd377aa4b : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ad2e0d9f5d14b72b5fbde7807d6af5f : SUCCESS in 1h 56m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b3a7425020b4626a708e410683251fd : TIMED_OUT in 3h 05m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c8a3cabed644b52894553df4e016ac2 : SUCCESS in 1h 51m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0c2926d8cb194ba5932a6d9df5677872 : SUCCESS in 1h 55m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d0a9a7f52b1c40d5bff5c309ad314de6 : SUCCESS in 37m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e507bdae05d042b986a69b23f784c50c : SUCCESS in 1h 13m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04166b94b02e4fe7893feec0baea58f1 : SUCCESS in 1h 51m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d953d7519f146ddbcadbfd46dcc2f0d : SUCCESS in 47m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e04580974b646218dea41c170ec1ef6 : SUCCESS in 1h 30m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/95999c52da8948a293c53eafcd919de1 : SUCCESS in 1h 39m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bca01bad2ae24695996594d6d7241ea6 : SUCCESS in 2h 10m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a0f1897c5f548a1ab9cad2fbe94501b : SUCCESS in 1h 50m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/724e28b4b5a6411da0266a2ecdc023c9 : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":5},{"id":"4a6e0add031e8cdbeb0c29882bceae08ee44e609","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-12-18 21:24:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/11/935511/5/check/fujitsu-eternus-dx-fc/3fdb2f4 : FAILURE in 2h 36m 54s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/11/935511/5/check/fujitsu-eternus-dx-iscsi/ed25db1 : FAILURE in 2h 30m 28s","accounts_in_message":[],"_revision_number":5},{"id":"146d977e6989934bc385bfae9a8261bad30d5287","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-19 10:00:11.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/5/fc : SUCCESS in 0h 59m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/5/iscsi : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":5},{"id":"28d496e8d3ff29e8aeb61b7ef1458c789b996e2f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-19 14:16:30.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/11/935511/5/check/DellEMC_PowerFlex_3.6/765b1c3/DellEMC_PowerFlex_3.6/2600 : SUCCESS in 2h 44m 06s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/11/935511/5/check/DellEMC_PowerFlex_4.x/ef61300/DellEMC_PowerFlex_4.x/1442 : FAILURE in 1h 54m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/11/935511/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/d6e49ff/DellEMC_PowerFlex_4.x_NVMeTCP/1312 : FAILURE in 5h 51m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"8402a88f844b943e8ba3c539a4e6687250ec287c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-23 21:07:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/5/20241224021207/NEC-V-ISCSI : SUCCESS in 3h 48m 59s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/5/20241224021207/NEC-V-FC : SUCCESS in 3h 55m 14s","accounts_in_message":[],"_revision_number":5},{"id":"80c191fe0cd99f0f8b7e32beada0041e6912c34d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-26 04:40:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/5/20241226054024/NEC-FC : SUCCESS in 5h 50m 21s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/5/20241226054024/NEC-ISCSI : SUCCESS in 7h 59m 44s","accounts_in_message":[],"_revision_number":5},{"id":"97fd3bc6b6bf35998fa9214f9a1c88cabe58f4ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 19:51:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_19_46_06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"cceceb3b221df669781c47f56ed09964e38917ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 21:00:24.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_20_54_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"fed62577bec3e4e7b3db89e02b9c5078362a6dd9","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-01-17 14:18:19.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"01d862fc8c109a3d0cdca5ac738b7ce3d0f6cee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2025-02-24 09:09:56.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"005f85fdf6792d4bfa18caa14311143da12569d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-24 09:10:06.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1353/","accounts_in_message":[],"_revision_number":6},{"id":"9d243bf9a27a6dd849c9bdb7fc9a35170938a21f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-24 09:10:09.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/935511/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"afac8c8c919ff1d9470168b372d1bf81cb432ba4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-24 09:54:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/935511 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":6},{"id":"7639a0f8f3d818cdf5ec6181bde1078bf826c7e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 10:00:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8702dea08fd4150b8c332242404d798\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e58d61368de54bc1b6d67b6ae1966740 : FAILURE in 29m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c203e063d144291abf3513b647ea7ee : SUCCESS in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4e67bc891ce719c775c434127cf61054cbe2533b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-24 10:18:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : FAILURE in 1h 06m 47s","accounts_in_message":[],"_revision_number":6},{"id":"7686780d695180d379170582d2b61ad2b7b0d33e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-24 10:27:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/380/.","accounts_in_message":[],"_revision_number":6},{"id":"0411a4de16e48ae68fcef07546d61256fef31297","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-24 10:32:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/490e7d44ce6243c096df5705ff457d43 : SUCCESS in 1h 18m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c4a6246f057142e782bdca1f274089a6 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":6},{"id":"c54c43dba35b9b839a0228dd5341eece0750dc34","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-24 10:38:34.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-935511-6/FC/ : UNSTABLE in 3m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-935511-6/iSCSI/ : FAILURE in 1h 28m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"7b0167b692ab387e48c1521e72f2c65b44cc1c45","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-24 11:16:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8427db79a4a64a288106e60e65395d7b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a26b6254ea54c21ace15e9f84ea2aa8 : SUCCESS in 41m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7584398ff279477186c58d2b0392fbb1 : SUCCESS in 58m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28fd0b58b7f54ba1ab3b99ef6bdc65fa : FAILURE in 2h 06m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a423eda2238467cb337bd7b1c286320 : SUCCESS in 41m 33s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8fe5be1477aa4b9892e5772a21cf7223 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/638d725288d44c1a8be7c1fbd624afb0 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/738bbb4b26e84573b0f2883420ff042a : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0bc5bd8e810247b8835acd556ceeee10 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"f231332391d18d617beb9c1ffb60741fb77bf4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 11:43:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3033834e65154ed6ab27bbb409696773\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276dd557cd98401980b8cb7c9b51e740 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1974c4d4270841a09ed8fa79cf2efaf0 : SUCCESS in 10m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8d1460cbab34ce7b8e6f116c289f2f1 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92da2c9c9dc54dd68d079913874a357b : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/855b24953bcf4d538efc7ba7c9dbfd1d : SUCCESS in 1h 00m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee907189ef374107bf6d0c9eb831e60a : SUCCESS in 49m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87df5580231b4df594254709b17e3676 : SUCCESS in 1h 43m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abd5356b2bcc4d80b2c5fdca8ac5147c : SUCCESS in 1h 11m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e21d205edc594a56a3b3f6b45bd1cb08 : SUCCESS in 25m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92468d2bb6cd4919b7be2da456f11cde : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9af5629d41704fe98198a01ae8d89596 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/04cc5b83f01c4a709ea8421c68098e96 : SUCCESS in 5m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c65af3cd799a4fb1a08a0fe3acb11b6e : SUCCESS in 8m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8368d99d544845db9b87be8fb9c00b89 : FAILURE in 34m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d6b052d47bd4a97a997f7674b0dd631 : SUCCESS in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a904db85fb32495d99d47fcb160bcd86 : SUCCESS in 1h 40m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ff933fba40d4ba98dd97f7c151650c3 : SUCCESS in 2h 30m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d7eeeff159d344c2afc112878df10511 : SUCCESS in 1h 39m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33724776e15348a7b6e5a6873ce16cfd : SUCCESS in 2h 09m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/86d0e4d159f745cca345ff8d64d6bc73 : SUCCESS in 35m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59c8617b32864cdeb5e013462155d216 : SUCCESS in 1h 09m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2402373ed1840b797512ee864aafeb6 : SUCCESS in 1h 44m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cb8bfe31eba74cce83fa8b3f1a4ddbc4 : SUCCESS in 49m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5eb81e3a0b844d5982f27861b5beb624 : POST_FAILURE in 2h 22m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1e44b85d66a8496eb4ec36818c0fbf41 : SUCCESS in 1h 55m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/affb3c30c9784a0dbd3b6a1e4102e92a : SUCCESS in 1h 01m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86de558a31c24db9afc494eb1a2a4fa5 : SUCCESS in 1h 46m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f3df0ce061644ec91b36fe7a60e5f48 : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":6},{"id":"aacc7e006873b1b493a41631a1964fcbe290fff4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-24 13:16:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/6/20250224180958/NEC-V-FC : SUCCESS in 4h 01m 49s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/6/20250224180958/NEC-V-ISCSI : SUCCESS in 4h 06m 36s","accounts_in_message":[],"_revision_number":6},{"id":"c3e3d2544f73084d9297e0b8307d6b50bb84777b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-25 21:45:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/6/20250226004922/NEC-ISCSI : SUCCESS in 5h 30m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/6/20250226004922/NEC-FC : SUCCESS in 5h 56m 29s","accounts_in_message":[],"_revision_number":6},{"id":"eb43a1dba16cfe656c0db8f1f104c7b7c634fefc","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-27 17:36:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"66d77d427f22865b77b0bfeac6c7ffc6b01fee70","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-28 06:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/6/fc : SUCCESS in 0h 56m 15s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/11/935511/6/iscsi : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":6},{"id":"5141c643d172ecb83a44b8d101cc3f08d635079f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"real_author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-12 07:25:23.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37477\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"}],"_revision_number":7},{"id":"d7eb0541c30e61fe9e03a9084e32f28a8f2a1916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 07:57:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35114a26ac24068be64134183925aa6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58ac4e446c83477d92e04043ddcc4bdb : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/512c4aa311f94a52ab4102329c0d08eb : SUCCESS in 24m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a4348ac026c48b483907f6b845476f3 : SUCCESS in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c7f9d5f57af08d2a9b7ae631c424917194ef0996","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-06-12 08:16:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"95a31dd22dd7445562dc802aa80428731066d851","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-12 08:59:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/935511 : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":7},{"id":"daa0db1797dfadc3e2d457ada3c4074ef2a16afa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 09:08:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/961/.","accounts_in_message":[],"_revision_number":7},{"id":"193a4b8306779bfea318a28d15b4e2b0800f7f58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-12 09:09:35.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"556487d6643321a4bfdbbf05e656f0f53a4f1e0c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 09:11:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1085/.","accounts_in_message":[],"_revision_number":7},{"id":"3a5e96e30a4277258d0ddbaa366c7fd138ae8bfa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-12 09:11:37.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-935511-7/FC/ : UNSTABLE in 2m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-935511-7/iSCSI/ : FAILURE in 1h 45m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"fdb8425d3934be0224d5536e2c532f67ba07c96c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-12 09:31:13.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ed352bda3a8d4c2a9d80b56dac6a8524\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bc9c914c210d454e87c8585a90361078 : NODE_FAILURE Node request 200-0000067024 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7d4c0aa489c4479bbba22a2d18e35bb : NODE_FAILURE Node request 200-0000067025 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a373e611bdf48a7b495bb41eaf62d6f : SUCCESS in 48m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce5eff30de7d49529015c7f16c6c0de9 : FAILURE in 1h 37m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dff996006bd640d18ff70bcd55f8f5b2 : SUCCESS in 45m 57s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40d1919c5dd54fb5a354aea0854a22b2 : SUCCESS in 47m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee93500fa5214967a4ad63f1b772c2a7 : NODE_FAILURE Node request 200-0000067030 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d35c737546834488ab85725f498885d7 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":7},{"id":"93742b09c6a92bc1dd72dc754428210aacb668e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 09:35:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/935bd8db67ca490ebab61fe9306ad449\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c2c4fada3b54856ab641e8345923890 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d9ed67c674143c7a5fbf5b59cc8b0b8 : SUCCESS in 12m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cd2f96ff07d4662b0f26d9124342163 : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5392eb331e547109a126eab2a1d79e8 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdac21decc545c69fb70134e04d93f9 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b532a831c5e444bab6e9f53807b61281 : FAILURE in 20m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad9a993236ae413d99aecd13b1b11f35 : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0426cf37d4340eb8d8c639d097bc092 : SUCCESS in 1h 24m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d0ab219ef54489fa9b2b12c8fd88347 : SUCCESS in 1h 05m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65167626fb3a411cb3d63ddacbbb979c : SUCCESS in 31m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64c7b82c6cbc483dbc6697fdd76efa73 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/477a9ae86e6f49f680675dec65aa4933 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58555ac7969845c3b58d08c6c68b5985 : SUCCESS in 7m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/86ffa649715946f3956bfda619e7c1d3 : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/866a97c85923415183e2a2b5a8fc2f8c : FAILURE in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4390f07c2a9645a2b4037efa3f09eb58 : SUCCESS in 8m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa3130479c4549d59e97af7d2b226ec4 : SUCCESS in 2h 04m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e98bea9f8d18455893d701ba6f574ae8 : FAILURE in 1h 46m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e414f5941154a21838ef8fa7f3fbbf5 : SUCCESS in 1h 44m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e0d7da194bd24363b4250880ff0d4795 : FAILURE in 19m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c378b57e5b5443be8f7ead1c17369641 : SUCCESS in 27m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4220867bdb94811a34921ff250dc244 : SUCCESS in 1h 12m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/35308c53fd61437795d2fb9c70271944 : SUCCESS in 1h 26m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8bfdb1025ce04ff3b7bece179cdfca60 : SUCCESS in 27m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/375792a8da4f445d8d69cc80caa873bf : SUCCESS in 1h 24m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7fabc4d19c9148b6bbcaa7547985d0ce : FAILURE in 13m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58733267a58a4b59946a3752b85e12ac : SUCCESS in 1h 26m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebd236fc0118464bac73868a37aadd58 : FAILURE in 1h 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db5c0b6c200e4235afdba37a64edc8cf : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":7},{"id":"17d7c8c9a36435c3811d48646dd6723e46b773f8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-12 10:03:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : FAILURE in 1h 21m 02s","accounts_in_message":[],"_revision_number":7},{"id":"b11f0b1dfe92b81ea0c3956a99ea8a6805036464","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-12 11:04:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4aa090076dde4ccf86310bc4669ec2a1 : SUCCESS in 1h 17m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ce6481c5f4ca451686b0d1366e569548 : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":7},{"id":"c30b4c6a55c3a7ad50717b8355301ca131ec4903","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-15 23:17:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/7/20250616033338/NEC-V-ISCSI : SUCCESS in 4h 23m 00s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/7/20250616033338/NEC-V-FC : SUCCESS in 4h 44m 03s","accounts_in_message":[],"_revision_number":7},{"id":"5458ebf641a489ab06350da109c098c84ae15471","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-16 17:13:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/7/20250616195626/NEC-ISCSI : SUCCESS in 5h 51m 02s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/935511/7/20250616195626/NEC-FC : SUCCESS in 6h 17m 10s","accounts_in_message":[],"_revision_number":7},{"id":"929b4a38a7ab0b0f02dd73dfdaaeef71d0c377d8","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2025-10-30 10:38:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bb9f79e549ad99b642109cf67eb1a926fc1700bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 11:29:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/020d9907ef814c5bae6c839da64d9bf3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea021ce321a547e4b891cd3feedc835e : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee36cf401c97417195a7e4f61ca90b87 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/901ac84d8c4446d99b0591580a44d181 : NODE_FAILURE Node(set) request 8272badfd6fe4346a7e17d365228c482 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e81ec131f8e19fa26f684446193dd2e4b655b3bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-30 11:38:06.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8149cf6e71b4f85971683c9843970ed : FAILURE in 46m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/84035ddd46174ce9bec53a185e73bd59 : FAILURE in 56m 09s","accounts_in_message":[],"_revision_number":7},{"id":"f8febbdbe841bed76c464f213aedc485467a2269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 13:15:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d7f07cc680b4fa4822bfdebf84062f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cadf234f4ed42abb7ef1892bc1fd809 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0e307420a364384929554def7d8e4db : SUCCESS in 7m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c16a9bcbaf54474f95132a7151c68f45 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064eb0addab84d939f3dfb4b2a842651 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0bf12973ed8c4acf8f3a7f704fd410c8 : SUCCESS in 29m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad7a32f9c0ab43deb44ddd84d9da86da : SUCCESS in 34m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2eb77aabf214b149d20373a0cd8bb8a : SUCCESS in 1h 15m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2432bf665ebe4e82b29a27604bb3f8d9 : SUCCESS in 1h 04m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c935d16719f04b29be4c120c1f840a6c : SUCCESS in 24m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2392cfc31877477db9670b4a5676ae46 : SUCCESS in 6m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/156f0a3c309a495b8ee70da68eeb0734 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d40803109854e5ca9caf6e8201d1d6a : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/341045b7545243e1aa25f983c695f95a : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e90148b7d26742db8389c70fdd4528de : FAILURE in 38m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35080f7fdde3481e93350a15c5bab052 : SUCCESS in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ddb591a67024e5cb555f14f2bd9a96f : SUCCESS in 1h 54m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40b377bc81984f15929a6b0c4da459ce : FAILURE in 2h 28m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5eb794a923be4942a6a78c7dd3294286 : SUCCESS in 2h 05m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ed91d28726d3446f80ee0afa14b67819 : SUCCESS in 2h 17m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/973e84737ce345419e5f57b1c5f3db0f : SUCCESS in 16m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c4b7c11ea9f499eb07444f274d20895 : SUCCESS in 33m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f48d98e37674deda193cb1df7f4e677 : SUCCESS in 1h 38m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/97f69f0c9fae4c2290e2e835082d2437 : SUCCESS in 42m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dd328865c44b4f85ac61ecc0eda78028 : SUCCESS in 1h 52m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1555a4d557a34f29a0d3bbf62f703118 : SUCCESS in 2h 09m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6fa0dab1912483a918db9a5abbd783f : SUCCESS in 1h 31m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8006b71130841969b4f3f60a0740992 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":7},{"id":"99d70d6ff47fd79f4a4ce76a48013685e16c695a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-30 13:22:14.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2025_13_15_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"a76a5c7f70773257e4dab7eef0e85e851be3cf91","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-04 15:29:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"95961c08490bb87ce19db4d669f0601ddb08d607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2025-11-10 16:37:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8adbb605cc116471239d99e3444c1bafdf5c079f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-10 16:38:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"5b508d04718ad7ed077f760c23cb2b14af7628a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:03:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c2e2204f234a03ad3a1325c294e972\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8fee373fb1244cbc8724b45b6d2ba84d : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bbb123ea64a44e89762705f12b60d72 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96485e42df4d4cf29a55bbc501d66697 : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9f245451ecf45a4ed6646111cccb7a04ab6ec4eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-10 17:34:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e32104a1755447c5bc5fcaf59d7cbb22 : SUCCESS in 43m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/975729b8a3104a72b83d2b0bfc9676e6 : RETRY_LIMIT in 9m 39s","accounts_in_message":[],"_revision_number":8},{"id":"57bba7580b61a8eed5ac7efd4cb7261a8971dcc8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-10 17:38:49.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I762dc886ff2b774a45a2798f011ed2d9cfa9937f .","accounts_in_message":[],"_revision_number":8},{"id":"cba13902829c07471eae681c40459c2a71865b72","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-10 18:38:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : FAILURE in 1h 13m 00s","accounts_in_message":[],"_revision_number":8},{"id":"9e19126e5e032cd911ebb316b2c7fed79fe68437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 19:14:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/007754ca7d6847a3b1b2975cc2879b55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93328d12b7b14447a4815f79b3874b6a : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e368dd0b1df24a60bf5c2a6371bf6a42 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88e66dbed8b746ba8acd6067b53e7d27 : SUCCESS in 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb53d321139b4d09aa2997730309cad2 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/61256a3b178647259e12391a2950d5ad : SUCCESS in 58m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4497112ec474d0d9ea4117d80fa2843 : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e6ce5260a004670996f2adea61bd4ff : SUCCESS in 1h 38m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd62b85ce7e94cf28c5df003dfb6d36a : SUCCESS in 44m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97c6e67fd6743d4b3ae6d2b74fbeca4 : SUCCESS in 19m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b66d826ed2843afb5c7aa8512f8f66d : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/29436145b3c5431cb30218dcf08fa8a5 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/907887e7138b48ed9d52870c2fddbd65 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a5bf53655b7460eb745ba7141293b2f : SUCCESS in 5m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/22cab437b3b44b3da777da3ede79f148 : SUCCESS in 10m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2616acabee334e49a23245d8f40057e7 : FAILURE in 34m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abc100845f7849d3a8f27079a2f10f37 : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eaab5672c2114551980dbcc9c65ae9b9 : SUCCESS in 1h 21m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6dd70f329c564acea0a6b57a6d75279a : FAILURE in 2h 31m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e58bcb51817440cc831092c1b009c796 : SUCCESS in 1h 35m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a34b279611fc4d82b4cbf4c9e2f93226 : SUCCESS in 1h 43m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a9db50ce51f453485c31c329619d134 : SUCCESS in 34m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb4691ad5c5648a6ad0a40c647c72c27 : SUCCESS in 1h 16m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/63a8eccc0f514bc58cf4a4fa93b9f486 : FAILURE in 1h 45m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9b1bdcea43964f4bacdece4e53117723 : SUCCESS in 38m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/239ab77affe24d82a41099315453c3fc : FAILURE in 1h 49m 09s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0fc31d421dd040f68ef4056824ce74fc : SUCCESS in 2h 09m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d51b0ea218243208fab5059362b5d25 : SUCCESS in 1h 46m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/728cdd062dec4f7780e980eeccbeeb0c : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":8},{"id":"0af3a2a9e52ddb3f3b35c12fbabd5c35648bf5c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 21:59:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2361/.","accounts_in_message":[],"_revision_number":8},{"id":"aa61fa80de31b140f5247ccc9a4f5b8e85f2096a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 22:31:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2475/.","accounts_in_message":[],"_revision_number":8},{"id":"0513fd5bf8a93b8c76155e153794413fbbcbe214","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-18 09:32:14.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fc5325f1b2339e0c70f38532fc6f6dee961ac1eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2025-12-12 15:54:32.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"96ab18325ae10cedf62a71b2c90327517f785f2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-12 16:14:02.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2847/.","accounts_in_message":[],"_revision_number":9},{"id":"b6c78b71132320616c288c8444f8f36a77728fe4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-12 16:14:10.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2700/.","accounts_in_message":[],"_revision_number":9},{"id":"bc0dcde3893a8003c49249a4c0a7d3394d098cb3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-12 16:52:02.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I762dc886ff2b774a45a2798f011ed2d9cfa9937f .","accounts_in_message":[],"_revision_number":9},{"id":"5bc97a0b27dbb9df69f5c2bec970827356eec829","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-12 17:17:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/935511 : FAILURE in 1h 21m 33s","accounts_in_message":[],"_revision_number":9},{"id":"27ee481f81224398b3d439a68a1fa31832961c9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 17:59:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd46ba54e62d47a1bd37955c9df572b1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/427f3ec5e8a64b99a7ab3335181477e9 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4c08fbdb750426aa3a99e15fb47515a : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2ba0a6ff5184a52a83d7f77de6b6cb5 : SUCCESS in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f46fe40aee34354c384462360ee393fff15042f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:30:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/800033c18442416b8e115a859ee77676\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196fadc2d92d41d58279a92a9416de1d : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f14f9dbb043e4659b9970e6d1b65e6bf : SUCCESS in 10m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/905426d854504bb9b1bbe96f04347b9a : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be2f496491445c4ac8d991e9df9caa1 : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/33ffed01e5734961af0de16877224744 : SUCCESS in 32m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19d525b4674d461faf4ec24b1108813e : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d310b6d5fa64e3598c1230af0e9ac23 : SUCCESS in 1h 30m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8804448a5e064d50bc859e5d5bd400db : SUCCESS in 55m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af354f2d681848ee8c88ef38a128c3e2 : SUCCESS in 15m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba007620048c48af972e2584a5c68034 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa1b1f30a0024801adb9ffe2bf0b8502 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9ebd9e738c94a08963f781439b29d91 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c23c08cc1724510bb2d52cb472f4d4a : SUCCESS in 7m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1664c6ac9b054a0fbd04f537dbecfe53 : SUCCESS in 9m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d9b4e5021b144749fe0d0431af8ab46 : FAILURE in 28m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a5d09785cd04fd1b041581d6543d8b6 : SUCCESS in 4m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c72c7c25a6c44ae89911e1f1e56e6b3a : SUCCESS in 1h 45m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8e710ed6a21467397500045848b3dfe : FAILURE in 2h 30m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d603c653852746e398621aab25682ccb : SUCCESS in 1h 49m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d0a8fe34f5514281a2636e4bda815b41 : SUCCESS in 1h 47m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9034159994b14d9e883c72a497360195 : SUCCESS in 40m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e0d18819bf64fca886409bf9a2ec4f8 : SUCCESS in 1h 09m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67cc9b8825974a30ac58a1969b981040 : SUCCESS in 1h 39m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5e7dfa1e6b184647bf7a66b61306074c : SUCCESS in 45m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79584cb962e64c27b21e9448b0f49072 : SUCCESS in 1h 56m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8ccb0069c0a64b1aa45601254c5beef7 : SUCCESS in 1h 09m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4590d3c07024303bf4dbf4787dd0c32 : SUCCESS in 1h 39m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/074ba68f7d174cd7851a654258acc757 : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":9},{"id":"23ccd158099604ac00e118aa46f7d3c8793b13c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-12 18:42:48.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_12_2025_18_36_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"82742bc88c55daabb10e2fa69e0facd268794f8a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-12 19:08:58.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ed5455ce7168424eb97736d4a5a15782 : SUCCESS in 1h 21m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/53ddf4832f2746f6b56395ad406bb260 : FAILURE in 1h 26m 18s","accounts_in_message":[],"_revision_number":9},{"id":"fdcdb1a9b03b1936b25e1e9e6c6bdc0ab6ad64f4","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-14 17:06:34.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-935511-9-fc : SUCCESS in 58m 11s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-935511-9-iscsi : SUCCESS in 1h 15m 57s","accounts_in_message":[],"_revision_number":9},{"id":"6c780e3049b44d3076929ecb531c348827652b56","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2025-12-15 14:04:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c5535308b2f4706365cba8e490eb8b24809ec862","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-15 14:07:26.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"452510a012cf8ce45c049d1d40371442c72c1af3","revisions":{"83314d5b8f9eab46606226f278e865be4f155010":{"kind":"REWORK","_number":1,"created":"2024-11-18 13:05:31.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/1"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83314d5b8f9eab46606226f278e865be4f155010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83314d5b8f9eab46606226f278e865be4f155010"}]},"branch":"refs/heads/master"},"259251dbbc33859e40b8916bee18705fe4b21509":{"kind":"REWORK","_number":2,"created":"2024-11-19 12:29:20.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/2"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-19 12:26:31.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/259251dbbc33859e40b8916bee18705fe4b21509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/259251dbbc33859e40b8916bee18705fe4b21509"}]},"branch":"refs/heads/master"},"7842716817883e986788a636dce84d53de06dd49":{"kind":"REWORK","_number":3,"created":"2024-11-19 13:47:29.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/3"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-19 13:47:08.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7842716817883e986788a636dce84d53de06dd49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7842716817883e986788a636dce84d53de06dd49"}]},"branch":"refs/heads/master"},"e6bbb82f09bcda68930a9faf7e305f2d8461ed77":{"kind":"REWORK","_number":4,"created":"2024-11-20 09:30:29.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/4"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-20 09:29:03.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6bbb82f09bcda68930a9faf7e305f2d8461ed77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6bbb82f09bcda68930a9faf7e305f2d8461ed77"}]},"branch":"refs/heads/master"},"ebc2369a4ce66cb0d55aa79ebfda903d236bc1c2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-12-18 13:18:12.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/5"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"Florian","email":"florian.doerheim@stackit.cloud","date":"2024-12-18 13:18:12.000000000","tz":0},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nRelated-Bug: https://bugs.launchpad.net/cinder/+bug/2092066\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebc2369a4ce66cb0d55aa79ebfda903d236bc1c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebc2369a4ce66cb0d55aa79ebfda903d236bc1c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b4ceb8fadf1e72469aad087d6d56edeb8cbb3386":{"kind":"REWORK","_number":6,"created":"2025-02-24 09:09:56.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/6"}}},"commit":{"parents":[{"commit":"bf3973eb52e5d1c4ef309d4938fa442c1cea076c","subject":"Merge \"Hitachi: fix to initialize a lock counter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf3973eb52e5d1c4ef309d4938fa442c1cea076c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2025-02-24 09:04:26.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nRelated-Bug: #2092066\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4ceb8fadf1e72469aad087d6d56edeb8cbb3386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4ceb8fadf1e72469aad087d6d56edeb8cbb3386"}]},"branch":"refs/heads/master"},"295d0a1bf6473c67ad504ff89ba435414f90a876":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-06-12 07:25:23.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"real_uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/11/935511/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/7"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"Florian","email":"florian.doerheim@stackit.cloud","date":"2025-06-12 07:25:23.000000000","tz":0},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the\nthe backup scheduling logic to give priority to those backup services\nthat currently have the fewest running backups in the \"creating\" state\n\nRelated-Bug: #2092066\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/295d0a1bf6473c67ad504ff89ba435414f90a876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/295d0a1bf6473c67ad504ff89ba435414f90a876"}]},"branch":"refs/heads/master","description":"Rebase"},"6eeebd69840b662d1d2bcd97f53f4bfd8e39e1aa":{"kind":"REWORK","_number":8,"created":"2025-11-10 16:37:48.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/8"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"Florian Dörheim","email":"Florian.Doerheim@stackit.cloud","date":"2025-11-10 16:36:51.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the backup scheduling logic\nto give priority to those backup services that currently\nhave the fewest running backups in the \"creating\" state\n\nRelated-Bug: #2092066\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\nSigned-off-by: Florian Dörheim \u003cFlorian.Doerheim@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6eeebd69840b662d1d2bcd97f53f4bfd8e39e1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6eeebd69840b662d1d2bcd97f53f4bfd8e39e1aa"}]},"branch":"refs/heads/master"},"452510a012cf8ce45c049d1d40371442c72c1af3":{"kind":"REWORK","_number":9,"created":"2025-12-12 15:54:32.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/11/935511/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/935511/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/935511/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/935511/9"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"FloD","email":"florian.doerheim@stackit.cloud","date":"2024-11-18 10:59:35.000000000","tz":60},"committer":{"name":"Florian Dörheim","email":"Florian.Doerheim@stackit.cloud","date":"2025-12-12 15:54:06.000000000","tz":60},"subject":"Add different backup scheduling logic option","message":"Add different backup scheduling logic option\n\nAllow the user to change the backup scheduling logic\nto give priority to those backup services that currently\nhave the fewest running backups in the \"creating\" state\n\nRelated-Bug: #2092066\nChange-Id: I762dc886ff2b774a45a2798f011ed2d9cfa9937f\nSigned-off-by: Florian Dörheim \u003cFlorian.Doerheim@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/452510a012cf8ce45c049d1d40371442c72c1af3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/452510a012cf8ce45c049d1d40371442c72c1af3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
