)]}'
{"id":"openstack%2Fcinder~937861","triplet_id":"openstack%2Fcinder~master~I11714210b5b350e750bc12fc50cafd07592b08c9","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"37535":{"account":{"_account_id":37535,"name":"Florent Le Lain","display_name":"flelain","email":"florent.le-lain@ovhcloud.com","username":"flelain"},"last_update":"2024-12-17 11:22:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I11714210b5b350e750bc12fc50cafd07592b08c9","subject":"Respond with HTTP 409 on resource conflict","status":"ABANDONED","created":"2024-12-17 11:22:00.000000000","updated":"2024-12-18 09:00:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f08f14d7976af6ee8ba331acc6692374797e709","_number":937861,"virtual_id_number":937861,"owner":{"_account_id":37535,"name":"Florent Le Lain","display_name":"flelain","email":"florent.le-lain@ovhcloud.com","username":"flelain"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 11:22:11.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-12-17 13:08:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-12-18 09:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"2e4106e0df1f412a819580e58e5f88ebbcf85b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37535,"name":"Florent Le Lain","display_name":"flelain","email":"florent.le-lain@ovhcloud.com","username":"flelain"},"date":"2024-12-17 11:22:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a231518f0771658ee34262f7271e1c93ee42dc73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-17 11:22:11.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1105/","accounts_in_message":[],"_revision_number":1},{"id":"d01a17f592bcf4ab425c7928a29a9e3df0bc6d6e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37535,"name":"Florent Le Lain","display_name":"flelain","email":"florent.le-lain@ovhcloud.com","username":"flelain"},"date":"2024-12-17 11:22:25.000000000","message":"Abandoned\n\nStill not the right ref","accounts_in_message":[],"_revision_number":1},{"id":"2f48378329be4051cdd80c4ae08200066704ed4b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-17 11:23:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/937861/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/937861/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"487758e206463ff6cc7214c9d0ce8b702daf0a9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-17 13:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/134/.","accounts_in_message":[],"_revision_number":1},{"id":"deba2ab3e44519053964d4ec4ec920966ee91379","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-17 14:26:29.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/126/.","accounts_in_message":[],"_revision_number":1},{"id":"6f08f14d7976af6ee8ba331acc6692374797e709","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-18 09:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/61/937861/1/fc : SUCCESS in 0h 57m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/61/937861/1/iscsi : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6579f4592b4c6ad338f65891448b24afaf87fdab","revisions":{"6579f4592b4c6ad338f65891448b24afaf87fdab":{"kind":"REWORK","_number":1,"created":"2024-12-17 11:22:00.000000000","uploader":{"_account_id":37535,"name":"Florent Le Lain","display_name":"flelain","email":"florent.le-lain@ovhcloud.com","username":"flelain"},"ref":"refs/changes/61/937861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/937861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/937861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/937861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/937861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/937861/1"}}},"commit":{"parents":[{"commit":"e953beffb6dddf608c0f37041f9ef95da1840418","subject":"HPE 3par iSCSI: getWsApiVersion now requires login","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e953beffb6dddf608c0f37041f9ef95da1840418"}]}],"author":{"name":"flelain","email":"florent.le-lain@ovhcloud.com","date":"2024-12-17 11:21:12.000000000","tz":60},"committer":{"name":"flelain","email":"florent.le-lain@ovhcloud.com","date":"2024-12-17 11:21:12.000000000","tz":60},"subject":"Respond with HTTP 409 on resource conflict","message":"Respond with HTTP 409 on resource conflict\n\nWhen raised, \u0027cinder.exception.InvalidVolume\u0027 with reason \u0027Invalid volume:\nduplicate connectors detected on volume\u0027 or \u0027volume state in error\u0027\ncomes along with a 500 HTTP error.\n\nThis change aims at returning a 409 HTTP error instead.\n\nIt takes over patch https://review.opendev.org/c/openstack/cinder/+/856041,\ntaking comments into account.\n\nChange-Id: I11714210b5b350e750bc12fc50cafd07592b08c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6579f4592b4c6ad338f65891448b24afaf87fdab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6579f4592b4c6ad338f65891448b24afaf87fdab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
