)]}'
{"id":"openstack%2Fcinder~938771","triplet_id":"openstack%2Fcinder~master~I7ffe7a7f08e7265250d46bd91f654d2a476b8958","project":"openstack/cinder","branch":"master","topic":"ibm","attention_set":{},"removed_from_attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"37888":{"account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"last_update":"2025-10-31 07:00:17.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-02-19 12:42:21.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"24005":{"account":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-03 10:29:01.000000000","reason":"removed on reply"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-04-09 11:15:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7ffe7a7f08e7265250d46bd91f654d2a476b8958","subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","status":"ABANDONED","created":"2025-01-09 07:08:19.000000000","updated":"2026-04-09 11:15:05.000000000","total_comment_count":37,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3834cf75b050a4ee64ee1f6bcbebe995eaa8af50","_number":938771,"virtual_id_number":938771,"owner":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"value":-1,"date":"2026-03-17 10:55:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 07:08:30.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-01-09 07:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 08:12:17.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-01-09 08:16:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 08:31:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-01-09 08:47:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-09 09:08:24.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2025-01-09 11:18:34.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-09 11:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-09 11:49:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 11:59:08.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 12:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2025-01-13 07:54:11.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-01-14 15:49:12.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 06:39:38.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-20 14:55:54.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-01-29 08:51:29.000000000","updated_by":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"reviewer":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"state":"REVIEWER"},{"updated":"2025-05-09 12:33:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-08 11:36:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-21 14:41:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-18 14:58:41.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2025-09-02 18:00:10.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-10-17 15:45:42.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-10-31 00:14:50.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2025-11-12 10:21:05.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-12 10:21:17.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REMOVED"},{"updated":"2025-11-20 06:53:25.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-11-21 14:14:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-01-05 14:00:27.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-14 19:50:13.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-14 20:16:39.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-14 22:15:33.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-28 07:00:25.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-02-05 10:25:04.000000000","updated_by":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"reviewer":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"state":"CC"},{"updated":"2026-02-19 15:02:07.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-19 15:13:07.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-20 15:50:51.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-06 15:30:26.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"}],"messages":[{"id":"55429d664d9b52fa846499e6ee6103c6d72745d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 07:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7f22bdf870888a304558f8887803d472e293918","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:08:30.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1170/","accounts_in_message":[],"_revision_number":1},{"id":"1e40867f9c5c4410d07be27c5071bff6de1f298a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:09:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"737a892297d8c4dbba5faa17008ab10c74545a2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 07:10:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a2e970428e76905477b323cde316a3ff5da05377","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:10:10.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1171/","accounts_in_message":[],"_revision_number":2},{"id":"8919139fd2942f07f121d8c93edb05ae1c8aa329","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:11:03.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"e6b277be157461a2ff73a04daee2f90851069461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 07:12:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"69a3589979ea4038ab45d7d34d875bf23fde7e4c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:12:50.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1172/","accounts_in_message":[],"_revision_number":3},{"id":"ef84465b53be0ec6d7efc6f3be50f6ac0b1f2e16","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 07:13:43.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"3a3b7f95210d52d1204d975c9d8a5a7f8047dc32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 07:26:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33b34ad05124539898a5323bd2a1148\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2dc5d89861f4c2f8391655152666607 : FAILURE in 11m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf090ae24b64427bb5e43bad7a3deed7 : FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2bd0b5e07be634d9c1cf9bacde203174214d8702","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-09 08:12:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":3},{"id":"0621b0d9b1fcea01f6dab5e1973dd81cbedbc57f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-09 08:16:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4e32bea4b917dbb2b291cd7bc205b3489a225d14","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-09 08:31:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/204/.","accounts_in_message":[],"_revision_number":3},{"id":"b1de7e6d67771d36595418395ac4beac4b3a769a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-09 08:47:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f45834638d6b4c5fbe613b85f58f49ac\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbfc469fbc77448f8154875af56f3b6b : SUCCESS in 38m 32s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a945b6f86978460182a17f4c6cf9fa70 : FAILURE in 44m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/019358846a1c42de81ae22d79447c472 : FAILURE in 1h 33m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c70613b01404f83bcab576b68907ae9 : SUCCESS in 36m 10s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/707dbe41413543ce93213c0f90162f5f : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84e3580ccf004b058bef18ec3f94c7aa : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f07c279c5080425fb817897cffcc465e : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/078c42fbc99b44d893a04a322afd6d87 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"84528c9f84eed6afcd4a058af380e161cd46d124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 08:53:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b77191aba30cf630ae60e8387c0536226c1a1c4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 08:53:15.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1173/","accounts_in_message":[],"_revision_number":4},{"id":"29c69666ece121469f7fbf78ab5de7fd557019dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:00:34.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"251bd9e7b8be2b7f240fe0c393bdbb6358b6c811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 09:03:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d5e773b96dacd0c4ed16aac6bf2ca56c0516461a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:03:50.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1174/","accounts_in_message":[],"_revision_number":5},{"id":"6015f3b3181c4519a8e58b078524cca56e4206ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 09:08:02.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a627fba030e3817857e662f4e1cd5ea4f5ea128c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:11:14.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"ee93a616a70927560bb088f94b6b5ac53ea5a785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-09 09:13:52.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"f770993061c03d078837c139070cf5a22f5cb1ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:23:40.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1176/","accounts_in_message":[],"_revision_number":6},{"id":"cd71b39ec127de1ff156018efd338d7320d2b6ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 09:29:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c9e52a4d00422d8fb48e33673461a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8bc3ee26cc34eb690d075e1c8f8f486 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c65e8a0b9d845f48d29081153c988f8 : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"073a85a04db4f00fd767ba7d34c3ec4bc0165e83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:30:39.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"3566e421c835ee3c4df65a7b94d17bb82cb1aba0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:30:41.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1177/","accounts_in_message":[],"_revision_number":7},{"id":"760e9f095ffc8797d5ff7b6da7662540d98ced03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-09 09:37:21.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"93f744f7a88912a783acbd0c7d24cda3c560db7a","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-09 10:38:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":7},{"id":"36e6bfe7adcbdbbe9ab84bb0b8e91d861088a541","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-09 10:41:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":7},{"id":"2d6b7acd65046c3ca87405ca24125fa28c01d189","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-09 11:11:26.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1b0d5e7451044e91a08fa910eef681f3\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3cd300446fd6460fabf3c53847f7dc66 : FAILURE in 38m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39e5d101703a432a88674c1737ca5db7 : SUCCESS in 41m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/771d474ed26849ddb4ebd70ad6c27523 : FAILURE in 1h 57m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fb7710ae59a4e92976585bae7cf4cd0 : SUCCESS in 38m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5612120d3684ef4a740b54e47e2500f : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38b5422fef2943ef950b7bdbf72f2dc7 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/417a86530c5a4979b1121275427a93ba : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57c7e01c969c4979bd8c1117d63e7e06 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"19fe64440d18ca3cac045222cde60bd0fe866659","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-09 11:18:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":7},{"id":"360cd4c7359ab139082d628b949698a789d68f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 11:33:03.000000000","message":"Patch Set 7: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d319a5bee55490f9c5552b761e439a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71008c97a00a4908a555a2a61fd572e7 : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b12cc59c581c461ebf1beab1e9a4af99 : FAILURE in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b6b7aa44ad148aa9f036c6fbf8331c4 : FAILURE in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0a49c089e841eeb8643aa2c114904c : FAILURE in 6m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6153dca59b804069b440c363b4154223 : SUCCESS in 35m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/218fab0010074796a6629fcc7c9b410a : SUCCESS in 57m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f10741ce6c914431843d8d8a7b28c47d : SUCCESS in 1h 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e5e5df341f247df86a719163effd402 : SUCCESS in 39m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/593c6c9e82e64e3db27c758ba40f10fe : FAILURE in 9m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/12433a771e0747d3b640c1b66a31ce3b : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b42cfb2d4dd146828e88055a93d5cf83 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d3a9501157fd42ba965acbc19cda86e3 : SUCCESS in 4m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e84068ab2bb842588c41404cfcb97397 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3aa14c533581412e8a1b0c82acecc142 : FAILURE in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf26bab091a54e4eb8b910ae9dc42afe : FAILURE in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fbe1783c5f64b12a78ceefbfcc703b2 : SUCCESS in 2h 10m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e14387c080714351ae17e5a7272f80fc : FAILURE in 1h 59m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02bd8843dcce4175b91db8e4aab367bc : SUCCESS in 1h 48m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/76228df0b6cc423682ff5321ffb7b295 : SUCCESS in 1h 38m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b218a7ec3c7147a8993bb6dad2fa401f : SUCCESS in 31m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/233da77526334c1190fb916f60cb0506 : SUCCESS in 1h 06m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fbd19b185b6d4b8eb1aa286124425431 : SUCCESS in 1h 27m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/707a3ec7f842440182b5f7635d580175 : SUCCESS in 39m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f47f1bb7ee7e4fbaaed413a3bec73356 : SUCCESS in 1h 27m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3d189f8638c54ef799b0dd952193b64c : SUCCESS in 1h 22m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fd68422e96d49bbad2578a8514982eb : SUCCESS in 1h 34m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/286aa6e53cd04329b9a29c650b4990be : SUCCESS in 1h 26m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4edc74bb83d248889ba56716e0708faf : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":7},{"id":"65f1e0c66ec495ae91c13ee3fe9642f27dc63c62","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-09 11:48:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/208/.","accounts_in_message":[],"_revision_number":7},{"id":"4cae987b5210234746580659b115c36868c0f938","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-09 11:49:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/66544e401df64f93917607210ffaf0c5 : SUCCESS in 1h 14m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e89f9df5b57744c4ae5f0ebbaf81e10b : SUCCESS in 1h 25m 47s","accounts_in_message":[],"_revision_number":7},{"id":"d15969bfb093dff7650514b1c7d7df18d63b46fc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-09 11:59:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/7/ : FAILURE in 1h 14m 10s","accounts_in_message":[],"_revision_number":7},{"id":"8da3cbe218992319b4fdeb77401a7e621318941e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-09 12:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/938771/3/fc : SUCCESS in 0h 55m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/938771/3/iscsi : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":3},{"id":"06daff7a3c282ec1338dfd88ddf9b506db1d3141","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-09 13:01:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/197/.","accounts_in_message":[],"_revision_number":7},{"id":"6ca802c07fbc51cdbe1c2fea7c5166eded32a586","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-13 07:54:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/7/20250113123033/NEC-V-FC : SUCCESS in 3h 58m 45s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/7/20250113123033/NEC-V-ISCSI : SUCCESS in 4h 23m 35s","accounts_in_message":[],"_revision_number":7},{"id":"99f07fa037c94f5f4e331d06473a57e3a53427f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-14 05:33:18.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"b85a7f96421d9a8db0696da6f0f2f3ca7ff33c47","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 05:38:31.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1203/","accounts_in_message":[],"_revision_number":8},{"id":"e1237c6d00c1f975d8e03b2d411457294c187b9b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 05:45:17.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"ac5dbfab8eeedb3fddb776ad9aee5d4d1da32632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 05:52:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c3b8083d8124224b9fcec234ef9889d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20eb2c0f06ad463d9329dc83869dc2f8 : FAILURE in 10m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c757ae54e2042859bef91a928207a23 : FAILURE in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"35830a94b9a1cc1376174cede6737851e3ecbaac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-14 06:03:08.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/227/.","accounts_in_message":[],"_revision_number":8},{"id":"ef1f2c903b62b636bd7094d0f94d73b32b639848","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-14 06:40:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":8},{"id":"2bbd2f84c0359170690eddf68ae7bdb653d52dd9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-14 06:43:40.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":8},{"id":"45b75670f5116b4c9c02be665f869e752f9e63da","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-14 07:06:38.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0d2cf953e9ba44b994957d845601df9f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f346ad9173ef4cb19fdecfa473af0164 : SUCCESS in 40m 36s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/855c7b142aba426a9416154a0532238b : SUCCESS in 40m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1aa9143b56434f2d8a69e2ebc70c273d : FAILURE in 1h 32m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0eede30d14534ee0a059f5390490c764 : FAILURE in 44m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a90358b8a6de4b0f93f7586ed15f9b97 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dfd033d1557c42e18f0bd8f3f34bc660 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da053671aa2f4d339d7910f4542f49c9 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0159201d19144ad6a7959f5222f87eee : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"e0985733b903b63903b975fb3a49889cdae9b9b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-14 07:20:40.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/216/.","accounts_in_message":[],"_revision_number":8},{"id":"83ac211da571c1bd811016a149772724d2536711","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-14 07:22:00.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":8},{"id":"6c52791245ceb3175904205f4e517b93da103c88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-14 08:00:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a43763750daf4d979813b3b9b44d0ec5 : SUCCESS in 1h 19m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5d226456f235488393e3625bc2f830b6 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":8},{"id":"863e0d3160f15cf80983d54e6ce8164ca7f48d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 08:10:37.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76c2a59b5c444758a8fd6a4bbeddfbe7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/375e5da38379484fac09c93a9ad52971 : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9689225c06714951ac8090c2e84108a1 : FAILURE in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41f38ab50f4d40b1919b6a35867da7de : FAILURE in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c49abb16214e7aa342e2ccb3515ead : FAILURE in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f2f9de9f0781488882f2c452649a292c : SUCCESS in 35m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bff1106e1a5452fb48aee1f0ad39746 : SUCCESS in 1h 00m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/182e6cd9ffc64ec49ceb2675ef9be8b9 : SUCCESS in 1h 38m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/daf7bf9b012d44eba165b548e1a1cc96 : SUCCESS in 54m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f2fac5e3b054780ad0cfb2c4f745cb5 : FAILURE in 5m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76ae7a2edd774469af604d157dadd55b : SUCCESS in 7m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0afac0f8be974ad4adf3d5c5b906819c : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7801b243ca3a46e482bf5ab265e0c29e : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/acb4259533d1438eb6e4a2e4af473077 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf140c7f135b463b87256019130b29b5 : FAILURE in 37m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c1558dfee7549df981629e3549fee1a : FAILURE in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71b78d020b794c55a3285d545a4c4a77 : FAILURE in 1h 56m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/db807c33d69d4a599002bb85873c69e4 : SUCCESS in 2h 23m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6396ca8668cf48a29b37a74d3f2cb7f0 : SUCCESS in 1h 25m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/760e51d5746749a9864cf144fdbc9200 : SUCCESS in 1h 46m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/852518cee78e4a42be2a07e8b407b819 : SUCCESS in 26m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c53e95d35f244b9b92405c9174f25ee : SUCCESS in 1h 02m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/644d7f2949c6434fa5d15c66e4a38acc : SUCCESS in 1h 40m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/685d11d40d82449b800dc0e0832ea4a1 : SUCCESS in 33m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9177dbe60ee54c4c979fd0087f12cca9 : SUCCESS in 54m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eac47a6b77c64a8287b1b345e0771acc : SUCCESS in 1h 43m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e8dda65066d4d49a2bfcee41e5087ed : SUCCESS in 2h 06m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd01791ce4264ed7a8072b18e7eb5207 : SUCCESS in 1h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a79ea46ac26f456585856df24ba5a005 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":8},{"id":"63443a1310fe8cb1ccac472f7462343ea59d404e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-14 08:19:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/8/ : FAILURE in 1h 37m 40s","accounts_in_message":[],"_revision_number":8},{"id":"704d05ef08ddb973c270a2501e9047956a3aec1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-14 09:16:09.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"887436589f4381410eedf32e08042069ffadd6ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 09:16:19.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1208/","accounts_in_message":[],"_revision_number":9},{"id":"db4fd2c1108aa603f46965ffd19e619d6e306cb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 09:28:39.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"cc423ca98dbea3933886afceb726a5140c91e9ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 09:36:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f86816eb6984ed6b879067cabbf1cb7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/974b8f7ff3cd4e23b05d9570e93151dc : FAILURE in 11m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d1cc83d2f914c90954cf4bf39768afb : FAILURE in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6b3f3458db839aecaf416876f763a17fefbaf026","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-14 10:01:18.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":9},{"id":"b2439487200274a2bde707a80b879956d7fd7c2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-14 10:09:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/231/.","accounts_in_message":[],"_revision_number":9},{"id":"bea9a24cb3f40660ef905a7e6288bb4228befa8e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-14 10:21:22.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":9},{"id":"5828b22c49b6905c97a7103f3e8aafe6f872a1ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-14 10:31:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/219/.","accounts_in_message":[],"_revision_number":9},{"id":"bfc436ee7768104c4c4d724f1befbac16ec115be","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-14 10:58:23.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":9},{"id":"c94cf86204dd344e5650706d9b3797994c00b7b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-14 11:16:55.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0a5a176479194d4082ee91145404ad33\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b998aee228c43b09a1c773a888d0208 : SUCCESS in 44m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6d8992ae4c5486292291eee62472c77 : FAILURE in 2h 00m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6251c3e8672f41569a92edac3c898cc8 : FAILURE in 1h 40m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a99d27bddf4a49a1874b8d57a80b5534 : SUCCESS in 47m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/deaa39794e0d48f0a2e4d85f23062621 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19ab003970584db398bd780b27ba55e6 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07093abb289a45c9b12f917023c2481f : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b4527dca948d4b80a0e3338cb4b89d03 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devs... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"0a3b8b664680f734a4e9152e99162083d6670834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-14 11:46:31.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"58a6ad6603b680d106a6440ee029d75bd291bad1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 11:46:51.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1210/","accounts_in_message":[],"_revision_number":10},{"id":"6fa1b9c45f6ca5262d14030307be52ced020cb10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-14 11:48:33.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"789136c15f0e3833b40dace723d4d64ccb58ca91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 11:59:57.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"faca82440b0f73372b6d53a259cb8052577d3416","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 11:59:59.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1211/","accounts_in_message":[],"_revision_number":11},{"id":"45a574bd8f4535404e659c7a4627039d33ac113a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-14 12:12:29.000000000","message":"Patch Set 11:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/11/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"c16e6fdd23e1dd97487db7fee6582a4ebac3ed94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:25:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8047b24a16040b3afbb8d7c964806df\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1a490e2c9c5455095fbe8b06dd6bc3a : FAILURE in 16m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/344617db640840dbb2d951ebe7b5c24d : FAILURE in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dae32d05cf1d697c3bdfc12575fb896fe16c1a19","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-14 12:44:38.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":11},{"id":"c6c4f9445e7e8c4792d19b8dc0423a6acfc7c477","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-14 12:49:02.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":11},{"id":"a82259342e4907d82e2c5f981fd861fb334da176","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-14 13:25:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":11},{"id":"751929eb4852666b8904c081e699ba751ce8d855","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-14 13:36:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/234/.","accounts_in_message":[],"_revision_number":11},{"id":"adee59870d79965915aa33e7739b891588ebee43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:25:16.000000000","message":"Patch Set 11: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9a023583ad049a891562a883c97fd76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a445e2c50b4bbc9ceaad986651e340 : FAILURE in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4986e11aa0554f90887c05c18e5b1d62 : FAILURE in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1afbde6f6db84c49888208f281d962f9 : FAILURE in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a9b2d1bca546d08896513c87ec0750 : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/0cdad4db19424732b7a28be2be8e69f9 : SUCCESS in 1h 00m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7d8a27d3e37449eb840611151484bdb : SUCCESS in 50m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d16d15baf614c55809c92283ef272f7 : TIMED_OUT in 2h 05m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3feeade074504ed484156078b1253e35 : SUCCESS in 1h 09m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a05c3898129a415a92df553722beb923 : FAILURE in 22m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0a2fc8e74f68403895844c7e9f4ed896 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/665e1caf5c2a4a0a903319b357a1642b : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d424a54048f24d90bcfb8fbe3c50c3bc : SUCCESS in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8503ac480a604af0942ff8a2c5d6e0b9 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a00c45e2abaa4beea21cc0555bb7fffd : FAILURE in 21m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7bdeb32c10ab4d0ba1150e37ba4222bf : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2706fb6f153245679de4949486414e06 : SUCCESS in 1h 10m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7181dbe97f0b45cb9602d284a37c1eb6 : FAILURE in 2h 30m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7b53b7234564dbc86faecb260ffa361 : SUCCESS in 1h 46m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ba7170dc9d6f49f282b58b46d628d23c : SUCCESS in 1h 50m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6edb03c3813f4fa2b12bf05a7a5a2b9d : SUCCESS in 25m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8b3cd466a1b4dad800b6fa4a7fca473 : SUCCESS in 56m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fbca0ab0ee654f55b8f0bb549b424ea8 : SUCCESS in 1h 40m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b2560399d6474a199a8d4e826ac6072d : SUCCESS in 47m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a5ee45bd875a4e0b9ae9c60487f488c1 : POST_FAILURE in 1h 55m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a48984e0f4af48f19b49fd69918e04d6 : SUCCESS in 1h 47m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0f2c74742ce4af3b06c605c61915f7b : SUCCESS in 2h 05m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e98949a04beb4db78b661aad61d50878 : SUCCESS in 1h 24m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cb318bdb21746a8a79154e53644feb0 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":11},{"id":"b655b088aa43f6c46c3032f3a7f88ca03acd5937","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-14 14:31:59.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d46ab293f37d451a968845d85513545c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/207d7ba9db574933927571cf7dc6df60 : SUCCESS in 38m 35s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3bcd7202407f4befa25517c649a0d8c1 : SUCCESS in 43m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa56c0447ec9495ea1c4745e58123a71 : FAILURE in 2h 40m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a1afc22d9984b988315a8f62d97657a : SUCCESS in 40m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bac30c2bc5bf4e8882ed7047ea2461c9 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d397e0689fe461a830167c601082025 : NODE_FAILURE Node request 200-0000022819 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9979e8f7d31f49bfa79adcafd0d87416 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eca4fe4e7f6a444d805d24ddddb4c871 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"e75e87ed646000972fc02f7dac8f6cec18aea056","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-14 14:50:18.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/11/ : FAILURE in 1h 11m 48s","accounts_in_message":[],"_revision_number":11},{"id":"6f0bfb6e60ac3b885a6e67497cccf3ff5b709bde","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-14 15:00:07.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/938771/11/fc : SUCCESS in 1h 00m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/938771/11/iscsi : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":11},{"id":"396c542b42502c1388dc49f3f5343549ce859f35","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-14 15:49:12.000000000","message":"Patch Set 11:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/71/938771/11/check/fujitsu-eternus-dx-fc/c0c59a2 : SUCCESS in 2h 42m 10s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/71/938771/11/check/fujitsu-eternus-dx-iscsi/4ddc5d5 : SUCCESS in 2h 31m 43s","accounts_in_message":[],"_revision_number":11},{"id":"b047b46cd469bcc1a1a30cd2ead024f748a4a319","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-14 15:59:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/89b52c6e2ac043139e3eb1505ef0162d : SUCCESS in 1h 16m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a707f67138bb49e4aa9f5668a694e08d : SUCCESS in 1h 17m 00s","accounts_in_message":[],"_revision_number":11},{"id":"59de7448d47c99ad1d1d9b79f1b7bdc6d447c5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-15 07:00:26.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f7c916da937e9fd2bbaa47d253ac9b7446cefdbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 07:00:37.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1216/","accounts_in_message":[],"_revision_number":12},{"id":"21e04c4bca362118c433c2f97fe2f93fe20e33dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 07:13:39.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/12/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"66c17af9a715a8e07e5cb6177de63a2e0d2407a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 07:21:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/258ea3c8870945fd876c598a21f668a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dffffdd957644a3b9137e723ceab10cc : FAILURE in 9m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/df8a103e16c14e67912f62b1a669fc6e : FAILURE in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8c53c4b893576ac1097eb3b5df7f41b0bf1a7c96","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-15 07:41:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":12},{"id":"cf053fe84e18e18d29c95d3dbb5e847bbcff042e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-15 07:44:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":12},{"id":"1a5fbc2ddcfaac299b55a31ea26bdc7eeadfc862","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 07:58:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/239/.","accounts_in_message":[],"_revision_number":12},{"id":"a7535eaed4074c50e48fb78582e639ec20e91a68","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-15 08:01:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":12},{"id":"9757d49679a3e89458e37ca3eef37359f6972ad8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 08:21:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/227/.","accounts_in_message":[],"_revision_number":12},{"id":"a3952257ca44d14b9ddca40f78d1c0434f770ef3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-15 08:24:31.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fec80b1bac6044d98c62b02f67165b46 : FAILURE in 1h 22m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3f24ff8683264521a24b835790b5799d : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":12},{"id":"5dcdf86e9781435955ab0836cddf5b209438effa","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-15 08:34:34.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/12/ : FAILURE in 1h 25m 34s","accounts_in_message":[],"_revision_number":12},{"id":"a792eba831f40e829975800da7ef699f29e18d74","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-15 08:34:37.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7692f668e70340bc9db89bd05fca3014\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9446d3a6fb374c66a85640a1e4a1eed7 : FAILURE in 39m 46s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d472b3c4ff744a3ab738f5f7f0d47ef3 : SUCCESS in 38m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fd60e947c55440f9a6b9754c3da7871 : FAILURE in 1h 33m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7cf51e2f9f244798881112d20e029ad : SUCCESS in 37m 14s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61b535de1d9d4e4d9a8b268982acdf12 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d3d15b4478904efea410d2c7dc4a2a62 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b83003fffb524e31a296622f03a38c5d : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6dea464b76c412ba1bfaf63d0b4b4b2 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"c5c692ced9187502f78f8e131eff09ced40d52db","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-15 09:46:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/71/938771/12/check/fujitsu-eternus-dx-fc/31fe6bd : SUCCESS in 2h 42m 35s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/71/938771/12/check/fujitsu-eternus-dx-iscsi/0001117 : FAILURE in 2h 45m 24s","accounts_in_message":[],"_revision_number":12},{"id":"ed42d56587ecbe698aadc9d80bbb02d5de331cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 10:14:30.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3e2807baac4c569339988fc7ec5216\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4e46ab663247a2bac03e9d201d6889 : FAILURE in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/778c9b0cfdab43f2a7643e5f8613381f : FAILURE in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ca290feb36747d9b2833341a157457c : FAILURE in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb05f4661af64bb1b52acbf1f2a08526 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/517d6443ca3a4f52809c67ba400c03fa : SUCCESS in 1h 09m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f367702c6b547899582400dbfc73abd : SUCCESS in 47m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/deb937324eb14c15a2900554d0e051f6 : SUCCESS in 1h 25m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e975f9444ce4d78a4fad672ac703bb2 : SUCCESS in 52m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7f8939a32d546a0a22ff0d3cff40019 : FAILURE in 21m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/321ba29306b74024bcd9897f774fe0a8 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59724a7110844f418846cf82fbea6287 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab2bbecc1c994e529920393442b07cf1 : SUCCESS in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/05531cd1a8434be4bee9df646ac2cf56 : SUCCESS in 8m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/042650b66ebb478392163d3003daadad : FAILURE in 22m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e35fa3fe585f410eace1a86ef9f8786c : SUCCESS in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd71295e41074534ac895ec379474806 : SUCCESS in 2h 02m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41c1795848bb446daeee5dbeccaefca6 : TIMED_OUT in 3h 06m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a44e84fce2940c3b24cc1322b7c16ac : SUCCESS in 1h 45m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a5d39190ec44457a97afe11092533296 : SUCCESS in 1h 55m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/246c02763ea04c0382abadc791b0928a : SUCCESS in 30m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1850e31f48f74e1e89f9458f1f95d6df : SUCCESS in 56m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47e49d9b604b4073bf862ea8d8a8fb4f : SUCCESS in 1h 34m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/999407f94f7044f795dcab3c5ca6b379 : SUCCESS in 48m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/248de5e03bc5453880a18099771c5e2f : SUCCESS in 1h 31m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b3147ac7cd6a46d0862832bc3e27066f : SUCCESS in 1h 44m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98464bf730ff4d4b856abc282fb126ce : SUCCESS in 1h 52m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7af9a8184ef745a08495a4db97358e77 : SUCCESS in 1h 20m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84e0408ba7024c39bcfdd0b47b08bc0b : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":12},{"id":"4ce766a163a5e7a3034554cf828ab36080763c0e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-16 12:58:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/12/20250116173501/NEC-V-ISCSI : SUCCESS in 4h 11m 52s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/12/20250116173501/NEC-V-FC : SUCCESS in 4h 23m 35s","accounts_in_message":[],"_revision_number":12},{"id":"0ff6ec6a6f0f0e1efa7a16ea06c25a7637db49bc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-18 06:39:38.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/12/20250118094807/NEC-ISCSI : SUCCESS in 5h 25m 08s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/12/20250118094807/NEC-FC : SUCCESS in 5h 51m 28s","accounts_in_message":[],"_revision_number":12},{"id":"b834c55b1a1120c1a16bc5b45f172a3756ebc798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-20 05:48:25.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"88cad79137f278246b88a7789e5f9c4cc72054eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-20 05:51:24.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"063197e889ebcdaf96455d68c80142d47ca35f30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 06:18:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc0fe486bf843e1ae9fcec1f3109528\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb6416e70963489c877f35359dfbbf6d : FAILURE in 15m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7be76e00fcdd4f5097da3776f17abf78 : FAILURE in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c367cfe49be92edac1c1e6bdba105e70a28a1a7f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-20 06:56:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":14},{"id":"815489d92090afb294ec46781e95610198520437","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-20 07:04:02.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":14},{"id":"f8f568ff2be1aea669b07b3cb310f83224653b1d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-20 07:17:15.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/14/ : FAILURE in 1h 21m 31s","accounts_in_message":[],"_revision_number":14},{"id":"0537b8877a90960e3f9c11344e911cbc54fba618","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-20 07:42:45.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/df61193f26a6406191171288b7fce696\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4bd1c2f7d31a4671995234e048dd6b90 : SUCCESS in 39m 12s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b6e3a7d9562438bb8a448f43cabe832 : FAILURE in 50m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cc14258e98ca4babb0405608584905b3 : FAILURE in 1h 50m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b2a2bdd13b25464f9a8db89230d70f90 : FAILURE in 33m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f39d4e752f74def8567c243f7afdfc1 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e34d5e8b42bf434cae1ce3a6572f28d5 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5633ccf5a32e402c9757f6305ebb81f6 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18c986184a564f78998af97b22880038 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"f8958881c6eb404df9308ab3b17751d40e14e6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 08:16:47.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59ea2781efd849a0bfedeb771925ed14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bdd8c38b634ec380a52e513bafbee7 : FAILURE in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d27eb7980e144c8085489dae6fda3223 : FAILURE in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c59a1cd86aa24fcfb92e6f40b7da2b2a : FAILURE in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f721d46667a543c9a84ef566d620d83e : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/72ef28144d234c508ff21bef4bc99991 : SUCCESS in 1h 04m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dde6dd3ba82e43d8b8c0e40176d909b8 : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4875f75d2cb4a89a07e933b324bf91b : SUCCESS in 1h 25m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74190104d2e74b80abf07598f1c053bc : FAILURE in 1h 02m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d61bbeb64abe4e30bfae27598968c412 : SUCCESS in 14m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/218addff517e4ffb994723bf8517b01b : FAILURE in 20m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/38eac0ba3a424f85b834258ee90c50c2 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/afee10623b8b4604bff6e52683bf1f2a : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/498ca1d2dfb344139221c0e408f1e4fa : SUCCESS in 4m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9abe073dc4a24cb0a4973942d0fdd148 : SUCCESS in 4m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/14fa239356874424b51a2979a91832ed : FAILURE in 18m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e00a13c47fc4fda90cca8b6988d4bf0 : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6ad23035c1c4cb7bd470b9a01a004b4 : SUCCESS in 1h 41m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9e1cb1392115454d9beccc4af70edd6f : SUCCESS in 1h 55m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/353a45e5aacf44e684a44eca91da3e3e : SUCCESS in 1h 59m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7bd01c7160b048a2ba08ee8d4be6daea : SUCCESS in 2h 19m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/674c132b217a4713ac57ea7d60085395 : SUCCESS in 41m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a42cde46e06844b8870c8e9ce8e3da8d : SUCCESS in 1h 10m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71e43f2bd6964eecae09c47de83e17ae : SUCCESS in 1h 21m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04d56d66383b48e9a55876982960feb5 : SUCCESS in 47m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b217d4129d0d4702828b8fc76cf88663 : SUCCESS in 1h 32m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/05c966c17c1f4630b27d4bfc6254efaa : SUCCESS in 1h 26m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77da3f5c6e8f48faaac387bf6c9a82e0 : SUCCESS in 1h 43m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6397079b60147458bcaccc2fa2d76fd : SUCCESS in 1h 25m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3a1762475474c03ac32f774638806ee : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":14},{"id":"2f6fa1ed0b697bd043d33453915cfd266536ac5f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-20 08:26:47.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/38bed2e195d647c9b5d7bd2e1f2ffd21 : SUCCESS in 1h 16m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1d34e81b28514f19b1ff577ff19a1b71 : SUCCESS in 1h 46m 56s","accounts_in_message":[],"_revision_number":14},{"id":"910a463c46808ee527b35d8821aef7cd1643fa72","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-20 10:04:23.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/13/20250120144826/NEC-V-ISCSI : SUCCESS in 3h 46m 28s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/13/20250120144826/NEC-V-FC : SUCCESS in 4h 15m 54s","accounts_in_message":[],"_revision_number":13},{"id":"26a611839673b0bf7892fe2c4607ed6ffd1433dd","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-20 11:18:03.000000000","message":"Patch Set 14:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/71/938771/14/check/fujitsu-eternus-dx-fc/3c0563e : FAILURE in 2h 50m 50s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/71/938771/14/check/fujitsu-eternus-dx-iscsi/0936847 : FAILURE in 2h 49m 33s","accounts_in_message":[],"_revision_number":14},{"id":"4c1959318365d380bb7b56e35296a9d4a352dfcd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-20 14:55:54.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-13/FC/ : FAILURE in 9h 7m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-13/iSCSI/ : SUCCESS in 3h 58m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"2a15000adb5dae614c7d25d8601912ba5c63b207","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-20 17:53:47.000000000","message":"Patch Set 14:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/14/20250120190426/NEC-V-FC : SUCCESS in 4h 03m 42s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/14/20250120190426/NEC-V-ISCSI : FAILURE in 7h 49m 18s","accounts_in_message":[],"_revision_number":14},{"id":"b471b7b1bceeb88c1a44203b50b00c658eb139da","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-21 13:14:31.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/14/20250121140205/NEC-FC : SUCCESS in 5h 47m 26s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/14/20250121140205/NEC-ISCSI : SUCCESS in 8h 12m 23s","accounts_in_message":[],"_revision_number":14},{"id":"c309a64712b547e7d4176cc4bccdce02236bd82a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-27 09:47:59.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"f5de4040515cfc92e3ba544cd4e531cb7b36d9b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 10:45:10.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265855e4b65e441699bfba9428e35d41\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb4500bdb2c448ae96029db023f80213 : FAILURE in 16m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/359f93b1961b4b5cb13088e0d90ffe58 : FAILURE in 17m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b07b160625f1ea0bf3f7ba1ab12e432a84cf3938","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-27 11:07:42.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":15},{"id":"619b04496a7bc7a4fec70274162e51301a6fe90e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-27 11:28:54.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"2fcf0ccdcb6910bd182b457e6a05c945e0f84723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 11:48:34.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff259445b2144507bd681a566fc832ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96523060000a40c384e856a10b01baee : FAILURE in 17m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d031659d1964248ae788dbde7655adc : FAILURE in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d45a3768ebab13f15d61c18b1ecf113ff6695dd2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-27 12:39:54.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":16},{"id":"aa5a9da0f9f2b4eca609e0f5c0a558c710d0b0cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-27 13:23:36.000000000","message":"Patch Set 16:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3629be0c2049463b88cbd55cee7b38f1\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a9b7ff2328b4b7abbf108330124335f : SUCCESS in 42m 41s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc41dfb2c66d4830908048f564957b36 : SUCCESS in 42m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e437d4c66a7543ffac78101766fd8e7f : FAILURE in 1h 54m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/10cd2f25ab76478f98cf4231d4b08ea6 : SUCCESS in 43m 33s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00b0c621307e4b959de413ea8d612e19 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/50f30bfa89d444e49e24410644273c6e : NODE_FAILURE Node request 200-0000024431 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/732ec81a4436492ebb831cc3bb0eac6a : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b626acd937746f09aa1b69b7d83e42f : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":16},{"id":"346085d658976a65800a531609d3bca5133d9b69","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-27 13:46:44.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/285/.","accounts_in_message":[],"_revision_number":16},{"id":"eba0f0592f0a99d840ebe2fc63cdc65c3f0d1b63","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-27 13:48:20.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 06m 02s","accounts_in_message":[],"_revision_number":16},{"id":"a75d378e558b6255dfaba947cb956d4d8c8ecb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 13:53:12.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bfc98b7e042449f8dbd7c3df12527b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57f2f4f0fed46f59bbabf30719af501 : FAILURE in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99b780bf6ef64d0dbf17c8ebeba91091 : FAILURE in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7afdd8c9071b4180b3e2966b6d7c6ab6 : FAILURE in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccfaaaefb254c5ba6b4cd2afe082b4b : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/ad37d4c4b6c042dfa7d2ed689585d1a1 : SUCCESS in 38m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af51d93eed0940bb981565d3e734fa95 : SUCCESS in 49m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/151e73c824cd4ce59a55677fe16d5b53 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/674a816b21a34fada8d1dd39c9f5e5f1 : SUCCESS in 49m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a239d100948e4dc99940cfb1a9b25c75 : SUCCESS in 17m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3bf990a957dc487a8ab92809bed47470 : FAILURE in 18m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0204a3db912e45ca938b454fbc7c4276 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6907623309144db69bebe6166b535c9f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df59c07e4a4b4d6bb1e077a27a7766d9 : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/29a7cb41996b49e3b65aefef5e04c8c6 : SUCCESS in 4m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a42f5f7785fa476bb55b0f65e6ab8cc8 : FAILURE in 25m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e1d0623888c4118aecd33b05ae2a9bd : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/208c58d908ef46ff9973ebb284ee8347 : FAILURE in 2h 02m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6db0fc57389c4ef782b4449b49b2a510 : FAILURE in 2h 13m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91ae34e269754ba6a7b91c115edddfee : SUCCESS in 1h 57m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0671abe49a29463b80a69825ad515929 : SUCCESS in 2h 13m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2518b335f48741aabe4c9104cc3e3db8 : SUCCESS in 29m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba8274f5c78841ee8c9a4d28af0c82e1 : SUCCESS in 1h 13m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c25b2058e184e42b800c9fd47dcfb11 : SUCCESS in 1h 40m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8166f8773c204f90a38d0f2dd7547df4 : SUCCESS in 37m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9847bc6f9b8b45eea8e4bcb3c0cd369e : SUCCESS in 1h 17m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/387ae3a1e7b944988809689c9aab83d1 : SUCCESS in 1h 34m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7225b4ff69a44308a8f2e014ade587bd : SUCCESS in 1h 53m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec306a4a9bdb46c7997524be2f93d8d4 : SUCCESS in 1h 42m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0663f8626c724346baa3d6226256bf54 : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":16},{"id":"24785f7517f5fa95d9231962fa0ad3cc1a1c3953","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-27 17:10:06.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/828fface2bad4954aed294a44636424d : SUCCESS in 1h 09m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1244376f7910402fb557a3400a552fbe : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":16},{"id":"1f8ab17e9b4ed91cd552cd01cf38b9d8f6431d40","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-27 18:42:10.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/16/ : FAILURE in 1h 27m 15s","accounts_in_message":[],"_revision_number":16},{"id":"bd54b2d6717bf44b6974b0cf8ddad55af8758aea","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-28 14:56:17.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/16/20250128180330/NEC-ISCSI : SUCCESS in 5h 25m 35s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/16/20250128180330/NEC-FC : SUCCESS in 5h 52m 43s","accounts_in_message":[],"_revision_number":16},{"id":"eb3c2e81840b5a6dd7590fd71b0d31c27002b406","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-29 06:43:38.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"e638fcaf45409befec84b983760f7a06c6782f56","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-29 06:57:30.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/316/.","accounts_in_message":[],"_revision_number":17},{"id":"4a5d086928af9117bfda6b06d7ca70032cc930dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 07:08:24.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9705c9f4234c60b8033c338f861267\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79cc52b67c9a4a2c9f4fd77c640b730b : FAILURE in 19m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de4f721bbed74c69a00d11513d2e2e74 : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0b4e5c490862fc8f00a8aadcf9be05bbea622fa0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-29 07:28:11.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":17},{"id":"0c60b92c896434a58c955a3497475964819f9792","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-29 07:51:04.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 05m 46s","accounts_in_message":[],"_revision_number":17},{"id":"da9572af9390bf89343b9cfc44b9d29df585ae49","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-29 08:05:33.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/297/.","accounts_in_message":[],"_revision_number":17},{"id":"0d755d2dc52ac45ea8a95256b713e590a1d70f1f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-29 08:07:50.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bd843ba686d048068d7749d7dfbd3826 : SUCCESS in 1h 10m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bf57137bade24efdba8067a7480d6cd0 : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":17},{"id":"b2bd795a2fa46777a6251baa2a036ca7926224d6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-29 08:22:45.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/17/ : FAILURE in 1h 30m 57s","accounts_in_message":[],"_revision_number":17},{"id":"21d4e954de44b08a73463e0f80b790f3d979749b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-29 08:25:05.000000000","message":"Patch Set 17:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0e4d8681178a40a19444c9eb9b0be106\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1e34758efdb74a79aa32de3341f5b725 : SUCCESS in 41m 50s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f4a6ec7fbdb405ea8ccdaa125b732ba : FAILURE in 41m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea10daceb54740cead9950a90b0992e0 : FAILURE in 1h 40m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3de485e0176e4987804a372bd1afa3cc : SUCCESS in 42m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ff98242a6374a59b3ad589a7b615744 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f4c45ff330964a66b244ef6478c752b8 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2fd550d068714122ac41d7e988da1259 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d044dd46e8943cb9c421970830a9f12 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":17},{"id":"b493e68005b6effb3a2fa41bb88f63ffe1a44b87","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2025-01-29 08:51:29.000000000","message":"Patch Set 17: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c09b0a1df054b2fa2f92304fce58963d4e151c94","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2025-01-29 09:25:34.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"79f58d690054d050791cb653477b2bcd9e14f5b2","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2025-01-29 09:37:08.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"75321f65e708d4737c8bfd76b165eab50937d54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 10:18:27.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d160df0096274b9a8cdee52e7734a76e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11064861bdce4f1d99f52a16af03d33c : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07897877c1a744d1bc915206c1fc3ce4 : FAILURE in 13m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f3a1d0d7d71401c9ff2607ce9402a22 : FAILURE in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53eded5aeb342e7844a6f772ced8030 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/fb6c6a2f157d46f194d465f9e5e11822 : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8cec012949b49d48523a395a872b655 : SUCCESS in 57m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55fac42eeb2f4a418f703d4680fc11fe : SUCCESS in 1h 22m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fd7859a47874d0290d12a3c349ee43f : SUCCESS in 1h 05m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91bebac4b0704888b735e99c2c9c530a : SUCCESS in 16m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e728677f88f4491ab7a6db662ed494b0 : TIMED_OUT in 41m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51e2cd152c41425ca262a4e8bc0dbf23 : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19102dd5818c4fd09fe7cc1abd4453f1 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60be1ac59c044b008cd7e7927254b247 : SUCCESS in 5m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/500b002889034774adb0cbb5cec15262 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd8aa3e142e94af19c28ecf5e0b58619 : FAILURE in 20m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6482257a47c46d0a6366faf285d2a6c : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8e2ce7e09924f1cadeb6d4174a5b2de : SUCCESS in 2h 01m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/89127e82dba7499a8122ab9363cf1bd0 : SUCCESS in 2h 33m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/712e40babb4b46a8820c52c355f2584b : SUCCESS in 1h 43m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0a7c5b81fe1f4ade94e287037504d495 : SUCCESS in 2h 12m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e30dd3b706004e059feeff4280c5c24b : SUCCESS in 29m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/903a8784b5354d22b9024100489b6bd8 : SUCCESS in 55m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f91810bcf62349fc9edb045afd713fea : SUCCESS in 1h 18m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9deca94e0f904e77a88914cdd10835fe : SUCCESS in 38m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a58c8d655b0d40b89e8dec04bf1d481b : TIMED_OUT in 3h 34m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d3c5e5337ddc452b8ac3c0e14993c65a : SUCCESS in 58m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/581f4746f261403ea0c5598ad9cbd505 : SUCCESS in 2h 01m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6146a31c11344f4a61ed1df3a08e9ab : SUCCESS in 1h 22m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6da70804482f4de7a9cc0c9fd6c08d4d : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":17},{"id":"5a4d0036234adea6ce982c597670c53522790a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-29 10:51:40.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"4b97ff61122c5bf6061cd395c3f3f848e47854b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-29 10:58:00.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"34221d61e5126a331aab37b0bab346b7f1ec9c73","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-01-29 10:59:13.000000000","message":"Patch Set 19:\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"16b6b8ab1edbf37c1308917fc12b9d00add87582","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-29 11:43:49.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":19},{"id":"a7363efe171d52d23cfbbbdd87a0b462f3c43caf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-29 12:07:45.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 05m 34s","accounts_in_message":[],"_revision_number":19},{"id":"c7a82bbcd630ce2f722c65dd49bcfaf841e3ead0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-29 12:17:15.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e29d4375662d4a1388ac25899a07cbb6 : SUCCESS in 1h 11m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/341f58bfd77749c9a82e48b84f08e899 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":19},{"id":"9d881faa8090ac3958b9403f6208f6854d4eef90","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-29 12:23:36.000000000","message":"Patch Set 19:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7c017877423c4c82af1021110eb248cc\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/54434468543c48e1857ca08214f545b7 : SUCCESS in 40m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1518c28ebcab42cfa0825c641e224570 : SUCCESS in 43m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f14cf8539f4645f7adcdf3b22228ce2a : FAILURE in 1h 25m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3eea4696794a456d8990a41a48fd46b6 : SUCCESS in 44m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a812c29ee91c4d89a348d93ed9aadf5c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/231a786fcb9340bc99d77194637792ac : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd301e19fab443cc9c317b4df68b4114 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/756fa612c0d7445faa0afa312e17cc7f : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":19},{"id":"f0cfb44f6fb0149a5f49504caee67dba1cd16017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 12:27:25.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f17146e6773b40568d2ff2ebe6d0b939\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5ff63a311e94c99b3b4b64f1473a52c : FAILURE in 22m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5333e471b2094c1a9ad78340ef43f174 : FAILURE in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b71faf848bdbda32b3daf3f3d852076641cd3a42","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-29 12:48:35.000000000","message":"Patch Set 19:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/318/.","accounts_in_message":[],"_revision_number":19},{"id":"ad6b3ead174dc476107bd1d048114ea8e67ffa51","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-29 12:54:06.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/19/ : FAILURE in 1h 44m 25s","accounts_in_message":[],"_revision_number":19},{"id":"d4635c3779c258de0d6777c03f62e3a96a2ac264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 13:36:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d908d9b9ca7464aa7f3371293956eeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a552f06ca84fbf9b54c9682cbd6d98 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04cbb2a6880948eab6abe31886f575fd : FAILURE in 9m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7484d8844ff24529bd184a2796070a31 : FAILURE in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68cdbfcb030846c5a8183d3e28d66933 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/cc8c9c54a29e4acf806aa2b4cfae0c7c : SUCCESS in 56m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57d7c4c8dc8f41679e474184e8287bd8 : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff9952822c654dce8d5f3f4af6e8b62d : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ac195e0d580438ab26c71230f6e33bc : SUCCESS in 1h 04m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87771351608c4b62b690cff87845761f : SUCCESS in 16m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d90d544e147e43ca93bb03d631426c55 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2cb8ec9eabf1478fb0648683cc619f3a : SUCCESS in 6m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b5d37d54aaf4ca0a8e2cf4e1226213e : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44013e755e2044528650ba07253d15e4 : SUCCESS in 8m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c9c67bcbe8a4d7db1962b9cff174a8d : SUCCESS in 9m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b04abc9d6a44c4d9d308dc6a2db4624 : FAILURE in 27m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe8ee88fc58e4beb885a5b8a3188b909 : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0899f2a0ea434b3fb28d45ea96ee4f23 : FAILURE in 1h 58m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ad881587fab49f5998e326452388aa4 : FAILURE in 2h 33m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e6cf21ac349447299332170094a0ace8 : SUCCESS in 1h 49m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/737f57dbd8e147ef830a2abf435589ce : SUCCESS in 2h 06m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c9c0e542268469f98e239f42a2f3d1b : SUCCESS in 30m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a1932a55261455f90e3b6928ad488a3 : SUCCESS in 48m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97ed8b5ef5e3402b9b8e812f5463633f : SUCCESS in 1h 26m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7fe880501c3d4f4cb5f92ef6c4f9bbfc : SUCCESS in 41m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/37f90d57000d45898f714b9631678eaa : SUCCESS in 1h 16m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2aaa69b8881a4d6d864b0c8bc19af518 : SUCCESS in 1h 23m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a5f29da06fe4424bd63ff0bad04c064 : SUCCESS in 1h 13m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62e71dc732ff4ed1ae4e75784fb6badf : SUCCESS in 1h 28m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c979a7aa51e4b62b67ac784eacc8c87 : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":19},{"id":"264533e6b2231478b4871253bbcd90c1124ba304","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-29 13:39:22.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/299/.","accounts_in_message":[],"_revision_number":19},{"id":"7d3493a840f99091bf2527427b66f22d5d25f713","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-30 20:21:55.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/19/20250130210521/NEC-FC : SUCCESS in 5h 55m 11s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/19/20250130210521/NEC-ISCSI : SUCCESS in 8h 16m 31s","accounts_in_message":[],"_revision_number":19},{"id":"e56640641efbdbc1053c5e8ff4f8cb0bbb58e76d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-31 09:25:06.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/19/20250131133717/NEC-V-ISCSI : SUCCESS in 4h 28m 29s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/19/20250131133717/NEC-V-FC : SUCCESS in 4h 47m 46s","accounts_in_message":[],"_revision_number":19},{"id":"273acb60ab3a7229f1a1cf4591ff9adc385cc826","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:38.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1243/","accounts_in_message":[],"_revision_number":13},{"id":"cf66d3b4985d303fc599f6f7bb31433cfa5ceb4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:39.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"9a87ddf72d39fa93076310a5ac78558d52a7c544","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:41.000000000","message":"Patch Set 14:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1244/","accounts_in_message":[],"_revision_number":14},{"id":"812e73eb09fdbe20c041aa38de36afe4ef665f6a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:43.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/14/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"fa3e4c9907590796a591d1c209f9530fbe5e4a77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:35.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1274/","accounts_in_message":[],"_revision_number":15},{"id":"12e65601a3455a77266b66740ec309c8b771cb01","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:37.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/15/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"1d8fa6c9403b6668c0f76b03d1544f1a6dfb311a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:41.000000000","message":"Patch Set 16:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1276/","accounts_in_message":[],"_revision_number":16},{"id":"d062951367f8cc257cfabee4eb69cc8db6455b51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:42.000000000","message":"Patch Set 16:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/16/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/16/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"bf62ca4ec616ec76141f6c0bfb1bec532c6f1fa4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:52:59.000000000","message":"Patch Set 17:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1282/","accounts_in_message":[],"_revision_number":17},{"id":"5acf82ac56007b91d5b000781f504aeb10eb31f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:00.000000000","message":"Patch Set 17:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/17/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/17/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"3f38c1e4dd54069255bd26035a1d497d404a613a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:01.000000000","message":"Patch Set 18:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1283/","accounts_in_message":[],"_revision_number":18},{"id":"b9f05326e00ce9a32006e596c80e246ac3f64805","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:03.000000000","message":"Patch Set 18:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/18/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/18/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"366926a438beb165af8a823f37e964ebe033234d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:04.000000000","message":"Patch Set 19:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1284/","accounts_in_message":[],"_revision_number":19},{"id":"6cf4c8e896649f43210e2ac9c21da2dbf7b251e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:06.000000000","message":"Patch Set 19:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/19/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/19/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"f23e0f81cb9345be68cba76d3e4b7a35c8b10dd5","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 22:07:36.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/938771 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":19},{"id":"471637af2e6a26165e21852111f7d66ffc7344e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-05-09 10:32:31.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"a196a5b762d9774eed3e6c06b41376fb1c741625","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 10:48:29.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac8866eae6024299840772cfcc2f4623\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca2bbd4720574610a24eacaee6adaf3d : FAILURE in 7m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c00b8f3e58034130a35a8c03881477c5 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ffe14a86d6048a0babf9dd42bf63186 : FAILURE in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4aaca84eea59fbc4d3224fd0460fcd6f91a48aca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-09 10:52:30.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/29a4cfe4a10441399df56a9549d7b84f : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aadd46c83e104d359c4ec89d8d0002d2 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":20},{"id":"c4bbfa8960fda7b41843925d68080e83d7f4ceac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-09 12:33:07.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"1c9e815c55c520ae5efa7f6c41f9c6d58d798f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 12:48:32.000000000","message":"Patch Set 20: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8f8b1fdbd3b4a45b58e843b4ccb4c26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/672adb58e7df477b9fb61328267c6466 : FAILURE in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ffcd99c777845569a00a4d611720468 : FAILURE in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a2feba5a24c48da907a49b306bb77d2 : FAILURE in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68f1ba4a7d4942fe90b41aab4d9057e1 : FAILURE in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626b066a11a146a6b6f472a4ec383f13 : FAILURE in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/39a13eb6936043edaee6be25ac10818b : SUCCESS in 1h 02m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/059b842225ad4243a87483169297572b : SUCCESS in 54m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b62a88d41424dcaac49037a863a94ad : SUCCESS in 1h 43m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/717e8aa019f9449cba753d1571a563f5 : SUCCESS in 59m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7d00fded9da4005b48fe2cefba20740 : SUCCESS in 18m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2244803a10ec4863bf88d9ee3863de8f : FAILURE in 6m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/223d6d9099864e9986663676ea92789b : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e449008b23ff428fbe16f430a93b7ece : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aab9f1f0e10949d0be805f28efc01c5f : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3c1212f854a4e48831cc55ccaf3c73b : SUCCESS in 3m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f91aa82c14124c688ffcf56467d4b301 : FAILURE in 34m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0649ebab084d486b8432643c572368b7 : FAILURE in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e15ce27b7a047d59210d1e6209bc176 : FAILURE in 1h 40m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5aeaad14168846888b37c1060e647faa : FAILURE in 2h 09m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d138849b0be84f53acb8c210f89403f1 : SUCCESS in 1h 50m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/01a8060930f64b1aa356f6b59254fab0 : FAILURE in 19m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5073f3d2a9804830a063e99daa33e203 : SUCCESS in 34m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee3b1195396c44a1ac4cfb41713f0662 : SUCCESS in 1h 04m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e428acf176544b29b4f3ca429c1d6237 : SUCCESS in 1h 20m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/10bfeef4d5d1433793fc0788564cbb89 : SUCCESS in 37m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf2de5b463ff478992aa14615e138e6f : SUCCESS in 1h 57m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/24e152e501b948208b8a058344975bfd : FAILURE in 23m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e2e986d455e43f28b4329e4732d12df : SUCCESS in 55m 25s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7fccbd067924bdb90fb0d13e2b26fa7 : SUCCESS in 1h 25m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb8d534e57647b196604282aa0577d3 : SUCCESS in 50m 48s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/volume/drivers/ibm/test_storwize_svc.py","accounts_in_message":[],"_revision_number":20},{"id":"d5dd0077832f7449724f395c56381de1d5955a36","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-09 12:59:44.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9b3245a349143ffae3ee7eba6d1b02a\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/404822c28e144e9aa8ae3f9cb8140208 : SUCCESS in 52m 37s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc0ae0b7519544d4aad4fe747dcac9da : NODE_FAILURE Node request 200-0000047647 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d8e499b7ad847c48c09387bb35f9912 : FAILURE in 2h 07m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f985d166c6824e50a700872862405ff1 : SUCCESS in 46m 53s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c3aceec369949c1a39ce9508cc7caca : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9118e4d29b67442cb458fce5be2c1b31 : NODE_FAILURE Node request 200-0000047651 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/73cb0f2ba3b14f2c9e2bfb0d0af4552c : FAILURE in 2h 12m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0f66d69678941a29dbb4389a4610e70 : RETRY_LIMIT in 35s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#68\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#82\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/d... (truncated)","accounts_in_message":[],"_revision_number":20},{"id":"561e67d40cd3936f87984e628ad4a7f8fb9874e3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-09 13:25:54.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 35m 10s","accounts_in_message":[],"_revision_number":20},{"id":"516620a46c665fcc708f467c5b5b5a74918a6b8f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-09 13:50:15.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 46s","accounts_in_message":[],"_revision_number":20},{"id":"525cfb6f1002bf873b4ae2b12bede33276472643","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-09 17:42:24.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-20/FC/ : FAILURE in 7h 9m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-20/iSCSI/ : UNSTABLE in 1m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":20},{"id":"937ee426c60725c2e8e078dbca685ad35408eb5c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-10 12:00:35.000000000","message":"Patch Set 20:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/20/20250510124059/NEC-V-ISCSI : SUCCESS in 4h 06m 30s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/20/20250510124059/NEC-V-FC : FAILURE in 8h 19m 32s","accounts_in_message":[],"_revision_number":20},{"id":"bb8bae7a7a12f7aee0b7af7480c033f8c21e84e1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-11 03:17:15.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/20/20250511035236/NEC-FC : SUCCESS in 5h 54m 01s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/20/20250511035236/NEC-ISCSI : SUCCESS in 8h 24m 36s","accounts_in_message":[],"_revision_number":20},{"id":"e52e2b55e36868f86b8e97351ce637f8ed36e831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-05-20 10:22:30.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"8c930e0a78ccaa1b219878b1a188e335df8e06f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:59:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464cb36073614b1da03e6c66f301eb1e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b226e4852e98430b8f5c4bc0151c699b : FAILURE in 12m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5b54b92711c430db3c699ba3affed58 : FAILURE in 21m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6103cf523e834301bdc22e831746c239 : FAILURE in 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"83d5a57227af19e92e025a02da7fa7f20e2f9eb0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 11:20:41.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/873/.","accounts_in_message":[],"_revision_number":21},{"id":"9417ae39b9eae2242ce649fb9db0bfc70a545e84","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 11:25:34.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/751/.","accounts_in_message":[],"_revision_number":21},{"id":"9398c5274ba465892ada4c4af38bb4136e394a33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-20 11:38:05.000000000","message":"Patch Set 21:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3f3cdb1da9d64d69b0450182cae774f7\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c3c4fa10d01c4b5c943db9d2245e6dd5 : NODE_FAILURE Node request 200-0000053378 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ccdd9849b8394ba09aa8bc3a40de1b0e : NODE_FAILURE Node request 200-0000053379 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2dacc0b2b75d49dda5ad51002d7b38da : SUCCESS in 46m 24s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90312b6c494a46e190f58c922e9271b9 : NODE_FAILURE Node request 200-0000053381 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3eaad0a18f1c4b5ba9d0851159393fba : RETRY_LIMIT Host unreachable in 4m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1438b5ec560d4834b3e7be9cb24e44b2 : SUCCESS in 46m 35s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab99e41885464e45a7f06bfe8af33640 : SUCCESS in 44m 20s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d4766db68cd4fe2b685852cab0d040b : NODE_FAILURE Node request 200-0000053385 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/49c8037230494adb9cffd3b820ab98bd : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a43d6731a6045128b267aa27ed56243 : SUCCESS in 49m 51s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job v... (truncated)","accounts_in_message":[],"_revision_number":21},{"id":"b50e40fc46e2eb913f649847a80ce80b89c7d347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 13:30:57.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e46b8b98cc0b408ab1e308dbbd953a1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e73038a634d480897eb5857356fa80d : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c122036f52ce401d97d52e25ce711bea : FAILURE in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3cbce4c0aab4a63b85ebf4dedce8b77 : FAILURE in 8m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0674ba8d045b4dcc8904800cfd87cefb : FAILURE in 15m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3434c63d594d9d8a4051aca8ed4833 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/da130a9cbe534d3ea5685c3c60149f0a : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94bbea1bb91b46468950c5a0b974864f : TIMED_OUT in 3h 02m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae11787408414422beaba77604c96cc8 : SUCCESS in 1h 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56d4f47465c44b99aa456aa617ab56e9 : SUCCESS in 1h 02m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd3088939ba40a0a358c762f57588ba : SUCCESS in 16m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7354a0efd3c9473194430396e2faa9ee : FAILURE in 27m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e82cbadb2e34be98dcfa69466a35f16 : SUCCESS in 6m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8f96fda6f9247dc8742434c618ab299 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d3a23a363c447cc9c28f2368a1a0924 : SUCCESS in 6m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fe3a8055447d49108f0638b607413271 : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d16166615014f909eac8cd216fd9f35 : FAILURE in 29m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06d48a29f56e489c9002f84fc76a1c09 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8184d5127f544eeb21627d82534e34f : SUCCESS in 2h 09m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/20597acab758495192b01eb7d4f567fc : FAILURE in 2h 43m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/730e94167023451ca529cf71a1865b8f : SUCCESS in 1h 46m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ab7ee1f9b69145e7a93838db64881073 : FAILURE in 17m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/18e9043ad2e54679a65c7e7ec7be7c3a : SUCCESS in 32m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e89176a5ca34b1d8fe92756315241b3 : SUCCESS in 35m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f92fbd246fba4257ae2f17b5cf052c6f : SUCCESS in 1h 26m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69db2916063c4340bdeec8683eeb2d2f : SUCCESS in 37m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b4c38ff66594d9c9d71299dd2a1de58 : POST_FAILURE in 2h 05m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e3932ed554c54417a8ff2270e96bcf72 : FAILURE in 21m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89bd1b9c4cd24d648fc7c174c10910ae : SUCCESS in 1h 52m 22s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43157e5530d6409fbb8971edc63337b6 : SUCCESS in 1h 24m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/636b1694785b4b2c8e1d682a5abf8228 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":21},{"id":"796877ca54e678acb885adf838c9d08cd9ee40f6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-20 13:32:24.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":21},{"id":"b3579355c7e118390515275bfb8fbb13f1da7cc5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-20 16:04:36.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 17m 50s","accounts_in_message":[],"_revision_number":21},{"id":"386150bc336edc4dfee4671adab4e231039d227c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 16:15:39.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"d4d6f6a2324d4a65e2cabae490ba9952da5fb8de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-20 18:21:36.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/87c88717a5554561bac39d12e20f2297 : SUCCESS in 1h 16m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e77afe1ca8704cfc9c54e0aefeaf24f7 : SUCCESS in 1h 11m 19s","accounts_in_message":[],"_revision_number":21},{"id":"5c647241bcecb605825f1eff163f9ff52018dac3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-21 07:01:18.000000000","message":"Patch Set 21:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/21/20250521063513/NEC-V-ISCSI : SUCCESS in 4h 14m 56s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/21/20250521063513/NEC-V-FC : FAILURE in 9h 26m 02s","accounts_in_message":[],"_revision_number":21},{"id":"c7823da82d026a38c4000459dc114b9d607a7fcc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-21 09:56:31.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/21/20250521100432/NEC-FC : SUCCESS in 6h 09m 49s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/21/20250521100432/NEC-ISCSI : SUCCESS in 8h 51m 56s","accounts_in_message":[],"_revision_number":21},{"id":"8feb14ecbd9e49a580ec59fb74bc9e527449b765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-02 11:06:54.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"d6ad14c8c55c95f4e64d62a5fd03b6a064e3698c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-02 11:45:40.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/ibm/storwize_svc/storwize_svc_common.py\n","accounts_in_message":[],"_revision_number":23},{"id":"a5ecfa3557b972bd8ea309c4a83858331a340c39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-02 11:48:47.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"9839f3d9233fc0f62b77e6f51489c6e1047b37bf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-02 11:48:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":24},{"id":"764b265f8a9963feffca2f7eaf602e83d9b3044e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 12:19:41.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c78c6f72c384729af3e694fff987a39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7152a4f1b6da4208a3170e377dc74ba6 : FAILURE in 22m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9952d881bb994ce5a9e1fdfb111ee259 : FAILURE in 23m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d5eb229f97f40168619d47385e0e9b8 : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"76292239b176ef2ff737d0aee73f364fdee1e1aa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-02 12:49:33.000000000","message":"Patch Set 22:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-22/FC/ : UNSTABLE in 4m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-22/iSCSI/ : FAILURE in 1h 42m 13s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":22},{"id":"e231ca5f6b2a647b5c6e657368579220066aa142","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-02 13:37:36.000000000","message":"Patch Set 24:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ea92342298c043d88b9834cb53a3a347\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f0f54eda651540279c332f5354d2ad20 : NODE_FAILURE Node request 200-0000057081 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/10d81fe682814706b33a621af40d9763 : NODE_FAILURE Node request 200-0000057082 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d50628053c0485d84c0c0a0e1d6e755 : FAILURE in 1h 05m 31s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/89fae12b9b9243758326365763d3848a : SUCCESS in 47m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1b02f938f484c93a26f2190f379605f : FAILURE in 1h 35m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2abac4fcc4e4fabb98b6ccbe92e714e : SUCCESS in 58m 30s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/acf54aed6afd4dcba8afe1c6f146399a : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/804cae21b960493fa926a79bb3da0441 : NODE_FAILURE Node request 200-0000057088 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3e3243b152848128b965c34771ed20f : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1c1ae3d7d694a8e97db58a3c030f4c4 : FAILURE in 46m 46s","accounts_in_message":[],"_revision_number":24},{"id":"4f6a915a486d49558994fb910a11a98293e7431e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-02 13:41:35.000000000","message":"Patch Set 23:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-23/FC/ : UNSTABLE in 2m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-23/iSCSI/ : FAILURE in 1h 55m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":23},{"id":"cc3552f5683ff69c9ec7cc5bb5cb8a5696847a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 14:20:54.000000000","message":"Patch Set 24: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f915bf52af7640baaa778ca8d18dbb64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77c474d5697460b829b15ea34b9c38e : FAILURE in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37ffdfc41e384861b644ca18b814e93d : FAILURE in 11m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ec8208235c5453f9d1f4baa2040c224 : FAILURE in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d4adc7bbfe043bdbc03839f64114e63 : TIMED_OUT in 41m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35f04fca78b4660b77cab636cf747e6 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/0e436f111f2b4dd5bc16b36dc48c1950 : SUCCESS in 53m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7fa953c9045f4d0f987d2cea077a62e2 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aed7368bb87241e0a9e3401805dbd395 : SUCCESS in 1h 32m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16ba0803a17d41619cd0643225e0573e : SUCCESS in 57m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dbf9fc658e341febf8360dfc9c5a399 : SUCCESS in 17m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9ffc120481e342fd8f2a0a9e7857e76a : FAILURE in 24m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d676892b506942c4bd0be3eb00f8053e : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df745bc21afa4d6ba27d1f53869f5de4 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a409ec6ebdd433b88b53fcb43c976ca : SUCCESS in 6m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f0b088a7048a4c17acf5a73006b6eeeb : SUCCESS in 8m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c7fe949ae85940d287154107d0ecedab : FAILURE in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c508e22ab2154d7c9f3692cf5d31d6fe : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98cb1e41d2304c5092503d290e3ecadc : SUCCESS in 1h 47m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/601ce8c3b80846978a4283b7365258da : FAILURE in 2h 31m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2782bdca77394520baa1e22702efc27a : SUCCESS in 1h 44m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/097a5b7a6693473991d2f2eb82856ed3 : FAILURE in 24m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e4f693ed141f4998ae402200f4140a3e : SUCCESS in 41m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/320deee21ce94c128b7652a892467e76 : SUCCESS in 1h 14m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d52e078ed3ce4a479d90102bf947bd05 : SUCCESS in 1h 11m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3960503c8da45f88d9278d56a36f55b : SUCCESS in 39m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21230ad7a10f42e284de8f752ff3e4fd : SUCCESS in 1h 20m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/55a4c3086815451489c5aa56c12a1ee7 : FAILURE in 25m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d82fbf4e5bf4a749c38c556a5d57b83 : SUCCESS in 1h 38m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13c013004ecb4098b79735f3f0f95e0c : SUCCESS in 1h 09m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b3ea63995f6499ea6c630c31542462a : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":24},{"id":"9a33b77cae73bb2774752a7e9d52feccae416c49","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-02 14:27:02.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":24},{"id":"7c22478e21c6d3485efa8a2568e7b3dfe11d9810","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-02 14:38:47.000000000","message":"Patch Set 24:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/808/.","accounts_in_message":[],"_revision_number":24},{"id":"bd042b5de3f9fa66095785b536d58ec5f1ed361a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-02 14:45:19.000000000","message":"Patch Set 24:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/932/.","accounts_in_message":[],"_revision_number":24},{"id":"ed3957375e519f66dfa0be206dbd3e14718e5e22","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-02 15:51:17.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"1c5ae8cc2eb029dd2db7cc6a90b98e5f351150d4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-02 15:55:52.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 22m 51s","accounts_in_message":[],"_revision_number":24},{"id":"c581a71090231624e9b0b8e072b2ec38facda196","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-02 16:48:26.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c04cc16ed0d14743bff9727188145b8e : SUCCESS in 1h 10m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2de4d841ed104ca88fe5e3e40c53db73 : SUCCESS in 1h 25m 33s","accounts_in_message":[],"_revision_number":24},{"id":"4ddd818273a9254e4926ad47ae9ec8d94ed0f106","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-02 17:36:05.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"4d508747fdf8594404afca012202defa4eec8846","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-02 17:51:57.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"1a3d82540f355dd79a33067d834fb55cfb21d6a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 06:29:50.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"8fdc99bb830d86d4ac00ff8a2f1ce215aa0956f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 06:38:28.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"5507cee438a81f80d426bcedd753857a5dee2ded","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 06:49:32.000000000","message":"Patch Set 25:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"e9c61941113240b014a4aefc2254d36078435e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 07:29:09.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2a8454ee60a451b94b3e0fc8cc85df9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b2cd7c92c5e47119a7714ee5c55bfa0 : FAILURE in 25m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9e087507b24b48c293da0a2bd9c334f2 : FAILURE in 29m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63ca100c7b7f40a29909e1908093075a : FAILURE in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"875250b3bf152acd850ed6f509c01e9b3b78b794","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-03 08:15:12.000000000","message":"Patch Set 26:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/db9a570287364ae396cbe64e0fda128f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69ca18e40a7943109ada44c3476c0e14 : NODE_FAILURE Node request 200-0000057509 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee3659ab8ea04f63be029cd8e150f377 : NODE_FAILURE Node request 200-0000057510 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66d8feff5ff44eb19874c8cef1d0db1a : FAILURE in 52m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d5ab407cb5c4f75a7e061703abc9b80 : NODE_FAILURE Node request 200-0000057512 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/58202cd2cf2543459f61f6a748f020e7 : FAILURE in 1h 35m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a671b77ca99f44129e14fd14232089cd : FAILURE in 58m 58s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2e4183b5f8e4a0680f22644fc196cf1 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb29ee94ee4547f7b0f8ed87245ba1b1 : NODE_FAILURE Node request 200-0000057522 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ab48af70c024f97ac84392803399e0d : RETRY_LIMIT in 1m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9949b82b5d3e424e9cc22f1b719f6711 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":26},{"id":"e3be1d512e262556d3ff2f89fa0463969d888f32","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-03 08:23:26.000000000","message":"Patch Set 26:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-26/FC/ : UNSTABLE in 2m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-26/iSCSI/ : FAILURE in 1h 44m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":26},{"id":"f7d0f66e5b45cdb7ff5e17cea0066639a49f6055","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 08:49:08.000000000","message":"Patch Set 26:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"a5fb242bfd55a05021908677b2bbf893d6236b34","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-03 08:54:32.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 37m 33s","accounts_in_message":[],"_revision_number":26},{"id":"813b5bd52c3425f75497511b682b90d878eda769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 08:59:51.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/430afd52378a4d8fb5870627c86b07fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b09333746940468655eb5d3d27f8d3 : SUCCESS in 7m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0515b20c5d1240619624c7a052325dc2 : FAILURE in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/430d0c4d67d84669ad95d1126559b620 : FAILURE in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57fb2708b52441e1ae83add90b7b74c9 : FAILURE in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8a197806c14310ae8d87592bde4e38 : SUCCESS in 15m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/53a8cf05942d481aa0043a15a65dc474 : SUCCESS in 1h 00m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/540e524f7feb4dd0b827ca9367a3c608 : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6eebbee2ce2b46edb8e8d02046f4a8b6 : SUCCESS in 1h 26m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1df1448b92b427ca090e97ae174fa81 : SUCCESS in 1h 05m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a752079fd2ab459cae9a146ea3ec035b : SUCCESS in 23m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0775cdce4ec541dfa1d80c498d7aaef3 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a612f5c8f6a408291a9c31fc932c584 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a757dd9348a24d8ead76780dd4672110 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/541e6b766a74466189f18ea41922a457 : SUCCESS in 5m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/729623b7874a4afd91ff4692d975abd8 : SUCCESS in 10m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d800319dbfe946398e0274e82d3cac64 : FAILURE in 34m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91018eb63f1b461fb36a878cad431c57 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/791b83c0ab324261be8943be5dcf9946 : SUCCESS in 2h 04m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85c8c3b9527847d1a8cb23ed57661f9f : FAILURE in 2h 19m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a880f6da95a941179c2b551425e3a45a : SUCCESS in 1h 36m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/51509ddf9c974fe8b5ee8b9bd9057b7c : FAILURE in 18m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4308d9bc06c945fd8d27deb7f5052f18 : SUCCESS in 34m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4fa007e7d21445b971196936d4b9de2 : SUCCESS in 1h 04m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/73e53c586060445fae7413080bca6f24 : SUCCESS in 1h 36m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54cb3ddbb137419fb1f07329fee19777 : SUCCESS in 40m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43f83fe72835400a97ea17004fac7d6f : SUCCESS in 1h 31m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/22e13b47f2cd4678949e8d0202c362bb : FAILURE in 24m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96f3603a01bc49d18566e54ee8e20f0b : POST_FAILURE in 1h 37m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc173dc6247d444ab825cee406922eaa : SUCCESS in 1h 32m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a985830e834a460f97a8ecfd92110235 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":26},{"id":"74eba55c37f24a6b29101e4957b15dfe0b6589f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 09:14:45.000000000","message":"Patch Set 26:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/817/.","accounts_in_message":[],"_revision_number":26},{"id":"8e500c35e5b883d5b2b0fb9a9a03e7d1774f1646","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-03 09:21:44.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 10s","accounts_in_message":[],"_revision_number":26},{"id":"d98ea77119d2f92d9a10b072036c123335c924e3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 09:40:16.000000000","message":"Patch Set 26:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/941/.","accounts_in_message":[],"_revision_number":26},{"id":"5d9b14a6f75155e8379c4aeacbee3a40a3ad837e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 10:18:03.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/ibm/storwize_svc/storwize_svc_common.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"ce1595a9e66f3b7b58858df3289c158febc910a9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 10:19:25.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"0f38e07680d3b084276359caa92e1fc7f3758337","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 10:19:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":28},{"id":"a3dd64cbfe4044c7feb342012437d4ccdd9c5a81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 11:03:45.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa1c0f73da0e42478e85da04081f3d37\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d664e1be22d74e76a4f12fe21be2c490 : FAILURE in 33m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94ae84f515a14f93aaf1c28f788cbf9b : FAILURE in 43m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5472031d24d47f4a730ee8807196b4f : FAILURE in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3a3acaab8d1d659e2b469acbf712ba5970553625","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-03 12:39:52.000000000","message":"Patch Set 28:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/caa199d76a234a5ab52c1c3fcc664123\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39d21f0982e24bab81806723b08bd924 : NODE_FAILURE Node request 200-0000057615 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/924f35a221dd4bfca47a522e90c9ee50 : NODE_FAILURE Node request 200-0000057616 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c75865467d5e49a08f8536884220210b : SUCCESS in 48m 12s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fed8263a29914c9aa270c6047d8c3bf2 : NODE_FAILURE Node request 200-0000057618 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86622b0e44a9447fb58ebadffa82eaba : FAILURE in 1h 52m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f1fce0a4d8d4a1e9fef046f7cb3cce8 : SUCCESS in 43m 25s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01eab55d00e44e60be230bc84edd620d : RETRY_LIMIT in 31s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0d2bbeed2c94fb49310a9d7aed62f35 : NODE_FAILURE Node request 200-0000057622 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6bb8475f6e64568877be4726f31ed57 : FAILURE in 2h 13m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3dfb2bea3789491594bd48647b7899c3 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":28},{"id":"25c92a33e3ff375ab28a61e507571c377e7ab59e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-03 12:52:17.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 20m 33s","accounts_in_message":[],"_revision_number":28},{"id":"360b681b753d2e74612938ef15eff83967651e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:52:44.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70776c3ac2ee4fca8b8ea851a5e8e69e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd0e499366e4f9ab52070512854d16b : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/967c6ff994ae4d9aad45d3127039d764 : FAILURE in 9m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b2ebf5fadba45179b922e6560f32604 : FAILURE in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5272bbce632b431fba00339300981e21 : FAILURE in 16m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9631b9f93a1c47899c39395cf86ffc09 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/72b0508e5bdd4fa3b7bf806ffbe971b6 : SUCCESS in 52m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67cb819829ba4b9faf9f0794e4c5a896 : SUCCESS in 57m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e52011165ea4a9580787cf30f6fbc1b : SUCCESS in 1h 31m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/507afa0039284ee89a208f9cc10ba108 : SUCCESS in 52m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2168ac98a13d4fdc8286dd16ba463a63 : SUCCESS in 17m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/148da5ab13a245a99875a01400eebf9f : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7e1c781e9a544680a192cb1b6009e725 : SUCCESS in 7m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2bd61b50f5c4eeda95549c9d7cd3ce7 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/368acb3a720342ae8d314f9cb4617b4a : SUCCESS in 6m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/61639b784fab4c6981fff6b75ab828c8 : SUCCESS in 10m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/022f01d373374efc8b23afd49417fb31 : FAILURE in 35m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50b6bc113d2f4e5dabdecc3f33301f0b : SUCCESS in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c1af20bf643418db1aa40752ba33646 : SUCCESS in 1h 49m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/79df17e36c2149bf85d3b42d8ab5903e : FAILURE in 2h 29m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b3b0c0a60164082a77bcbc5e5947686 : SUCCESS in 1h 57m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/85f5d48740f747459a5d630c9fca4f87 : FAILURE in 23m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c101839c0a74fa9b793d1bb19d2dae6 : SUCCESS in 34m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c97a688ed59e44789be638978c9c20a7 : SUCCESS in 1h 09m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92c6c8571c7a47f98342b089b5265f9d : SUCCESS in 1h 32m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9f66540bc0aa4d129fb2a5027f07b1f3 : SUCCESS in 37m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c7407278960749dca66dd909647c3aed : POST_FAILURE in 2h 01m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c47b20fb5329417c8a66a0b821099afe : FAILURE in 18m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34e54f5048a840cc87c1bab972f1463e : SUCCESS in 1h 52m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6ee1e684c2942479a1758e9b9bc1ad8 : SUCCESS in 1h 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8765a075c6504fefb387c1c7ba6a1174 : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":28},{"id":"25fb696e8438d55d19ee177a81b6b093ec8d4e20","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-03 12:57:20.000000000","message":"Patch Set 24:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/24/20250603122938/NEC-V-ISCSI : SUCCESS in 4h 19m 47s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/24/20250603122938/NEC-V-FC : FAILURE in 9h 27m 39s","accounts_in_message":[],"_revision_number":24},{"id":"d7c8ca9b9aa388f3030f46bd93e7a96266396c2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 13:45:19.000000000","message":"Patch Set 28:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/945/.","accounts_in_message":[],"_revision_number":28},{"id":"cf8c710a2441fe2f28da9542523b46c1149bc6c6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 13:45:55.000000000","message":"Patch Set 28:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/821/.","accounts_in_message":[],"_revision_number":28},{"id":"438095ecd3b15409fa884ebbc27c443728888b08","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-03 14:19:32.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/53629a01e9474522b180e13487bd634c : SUCCESS in 1h 19m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7481181d35d34cd7b65ed90403b452e6 : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":28},{"id":"367a8f780f52804fa45685fbfd1bbc28dd5fddb2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-03 14:47:24.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 19m 01s","accounts_in_message":[],"_revision_number":28},{"id":"a89649d97b41ecfe2b4dc460151a69b3f7ced8c9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 15:28:32.000000000","message":"Patch Set 28:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"b7885d5035236e18f42d76f15d765625304bb70e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 16:27:58.000000000","message":"Patch Set 27:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":27},{"id":"9e76e65291f5b31d388c9bfdd88970f27d2ac9c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 16:54:13.000000000","message":"Patch Set 29: Patch Set 28 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"36740b6c5c8cf7b874b136531dae8288497b41fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-03 17:18:11.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"b2469f93dac4d367550e4a36deed7f243ed04146","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 18:16:33.000000000","message":"Patch Set 30:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/832/.","accounts_in_message":[],"_revision_number":30},{"id":"19e8d0104e6859b51ff5266084c7504eb6f75a99","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-03 18:16:52.000000000","message":"Patch Set 30:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/956/.","accounts_in_message":[],"_revision_number":30},{"id":"0112114a4805da64226367c40abbaa1cadc2b6ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-03 19:05:07.000000000","message":"Patch Set 29:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-29/FC/ : UNSTABLE in 2m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-29/iSCSI/ : FAILURE in 2h 10m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":29},{"id":"7f2d63fb7fedee33967e1409eb3f436ada956319","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-03 19:17:18.000000000","message":"Patch Set 30:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-30/FC/ : UNSTABLE in 2m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-30/iSCSI/ : FAILURE in 1h 58m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":30},{"id":"ef942426bedeed4402d3a338dba4ab2eea5d8d80","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 19:47:33.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":29},{"id":"09fac9e81c189af807ce3c588a26c4ec8f157f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:02:34.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e1956a9351416a98ac28b7ec360592\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb48e7a78bf8407ba9c41e5c2689a064 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7298d7027ae444fe9d938b965163bdf8 : FAILURE in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07d7017916e9427a85c31e7bfd92bdaa : FAILURE in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ce5c62a1c1d4ad7b5492f5042a4dea8 : FAILURE in 15m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11e01ab7e314b8eb2523b8c5a495652 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/2f656a3d083a45e18d02d60e1f2626c5 : SUCCESS in 1h 00m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8778e002b17d4558b00ef8edff8020ac : TIMED_OUT in 3h 02m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6052334fb9914222b7136b2bc4975725 : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a011af0dbf674cb98f76ee542ddb92d9 : SUCCESS in 1h 01m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7172a0ecda7434eb746a573094acb4c : SUCCESS in 21m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/509d91477cd1403ea683ce920e78e19b : FAILURE in 26m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ba6e3fc7a2a478488149491d2d4edac : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aef9234a4e804801ba82824c4c6e6395 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ab69192ff77946dfb5e4ac5419356b6a : SUCCESS in 5m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/86d0ee2884ae42feb5fac7244621df3c : SUCCESS in 10m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d26378d3e846427bb2010e6c8e647242 : FAILURE in 34m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7817fbbb04594a7a8c9d5cc0232e4a6b : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14de42dc99b84b28b88b5a3ed8bab933 : SUCCESS in 1h 53m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34e30eb0f32c433f8a17ab2976e173c6 : FAILURE in 2h 41m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/43e3bce6c23846d5827e6db5fe4971b4 : SUCCESS in 1h 56m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/888c8b2057d446afb69bc7804003b658 : FAILURE in 21m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1c8c6f23765442ddb3a582da201d9c5c : SUCCESS in 29m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b3e06f7a0df40f0a40192668f8bcb53 : SUCCESS in 59m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c0a329fc4204e0aa2c24122f14aaff2 : SUCCESS in 1h 20m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/865b382aa827442a9aeb7d8a0a5fad3d : SUCCESS in 42m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a6a065e5a2814fd5a0cb09efe3a4b731 : SUCCESS in 1h 17m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eae54f0acabf436fba207a05ade13217 : FAILURE in 24m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b88d536225174e92909af5276e62a9b6 : SUCCESS in 1h 30m 37s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c598b2d4859d4bc093bf9e533f7e2f08 : SUCCESS in 1h 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b44694fd874c4abd4820c71af1fd00 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":30},{"id":"b0e093bc1e744a00d90e7dfcc9212d5c1677c25f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-03 21:09:49.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":30},{"id":"f8c34e049341aeb02c4bcc9aad80101d808d46d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-03 23:06:50.000000000","message":"Patch Set 30:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"9318fd232115247ab7ba850f5270d6d57406045f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 23:45:14.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/633c2b60403a4f5ba0605274f5edd7b2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af189e321de74662ba08afbf505b7c9e : FAILURE in 56m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0965b65fe22e487fa9fe73c2dcf67373 : FAILURE in 34m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b6d6b2a0f1fa4d4388565a61fbea57a1 : FAILURE in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"547bc986c7e49e1a6c855b3a70ce9cc400c4c67f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-04 01:09:05.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 19s","accounts_in_message":[],"_revision_number":30},{"id":"e39a7514885f265cb048218984b16a6f1a57c6f9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-04 02:33:56.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6ab621b6babe476e90f35b3771eec9ee : SUCCESS in 1h 19m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cc61b95d775d42e49e243702ac39a56d : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":30},{"id":"de004ac9d8745936805627d0336e7497d3de2c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-04 06:18:30.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"fb1f978fbcc2821ecc0c1bc9ed6bf8f6e605ac06","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 06:53:42.000000000","message":"Patch Set 31:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/839/.","accounts_in_message":[],"_revision_number":31},{"id":"c08abaf4bffb52b04fe9b81cd5af271efccb5e47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 06:57:08.000000000","message":"Patch Set 31:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/963/.","accounts_in_message":[],"_revision_number":31},{"id":"7dde1a3cde088c3c932cdc7439a30703432f3362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-04 07:56:06.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"c1e29f4afa625df253732867c5435f1a5408d535","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-04 08:08:29.000000000","message":"Patch Set 31:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-31/FC/ : UNSTABLE in 1m 57s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-31/iSCSI/ : FAILURE in 1h 49m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":31},{"id":"065e428f08687a6753c9666ae55722a7f4116537","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-04 08:14:30.000000000","message":"Patch Set 31:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":31},{"id":"a23258f57b4e7b006de08241b031b036974b4abf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-04 09:02:33.000000000","message":"Patch Set 32:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":32},{"id":"89e560e89cd5ce4b524f10f4aaf0a7e94c120a30","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 09:03:06.000000000","message":"Patch Set 32:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/842/.","accounts_in_message":[],"_revision_number":32},{"id":"e8d1c03df7bc4cea434fedafef645c0da680884b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 09:06:06.000000000","message":"Patch Set 32:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/966/.","accounts_in_message":[],"_revision_number":32},{"id":"378427a59edf31583b50c95ddc836b75722df413","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-04 09:44:06.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":32},{"id":"293b8931fe6c344c22fbe1ba0d7f37141b9c1169","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-04 10:26:37.000000000","message":"Patch Set 32:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cb675297a2c3496b976e7fe066b9a283\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a28e29b74b84f8488236cd7b8a51e7a : NODE_FAILURE Node request 200-0000059059 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc38ac8f2eed439ebef6c3f0ef18152e : NODE_FAILURE Node request 200-0000059060 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7ff07392f0204e4d850462027ffa77ee : SUCCESS in 50m 04s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be583ffe26844f09aa481d014594e970 : SUCCESS in 45m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/018e7e8d6fff489095c6958b37965859 : FAILURE in 1h 37m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52f670cf664c415e9719404aef3b7ed0 : SUCCESS in 47m 47s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1b11e5d101449eb9bb7ca28f230f768 : SUCCESS in 56m 55s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/63627e89b14a4b029b2760517c21c531 : SUCCESS in 1h 05m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64b510f1f4b5456fa37ff9b1b4710877 : FAILURE in 2h 29m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8dcbf94a4d843b2b64d05790acd4eb6 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":32},{"id":"f51d6225d275c4e8ebd21092aa890f48216b61d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-04 11:27:27.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":32},{"id":"1a23ea4251e160a8fa926ea4c942dcf6ef37f5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 11:34:48.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9707229fb30a42c4bdca244dea0f035c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51157be2df764ee58c2f28fbdc8e1984 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fd4d7276d6a4ef6a530e7aa78490243 : FAILURE in 11m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f31bee073ea4ec88b2da86937cde494 : FAILURE in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0de1d15aceff411c8c3af8b5b8b5eef4 : FAILURE in 13m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33293e2637a4db886b7a09345ca768a : SUCCESS in 16m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/4287dbcdf2ce455c936d6f2a988ef78a : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eaf9406fa4bf4bc59898f449af008a88 : TIMED_OUT in 3h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2cdd13f9f9ce4fc98c90c9f1089ad65e : SUCCESS in 1h 29m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5023f0e22c44587b15be6e2e7490cac : SUCCESS in 58m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a182a4f5ce74412b8c15da61a7dac8b : SUCCESS in 18m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c1358bfd46c40598c6dfcd161a47d11 : FAILURE in 23m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/80dd3dfba3364725ab40c7f2fcaea259 : SUCCESS in 7m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/490bdace34c944d1a258ead8779eb636 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/115ee32b2ae546b7951b090907836fce : SUCCESS in 6m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b464406a08e4abfb60a7cc9342981c6 : SUCCESS in 10m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3156e9e2989b44b48e6c1a86ddc279cf : FAILURE in 38m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c9fb1a877d44ad4a6cdd903a1b59c0b : SUCCESS in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a7622c98271d4398b55a1f48bac4f0b9 : SUCCESS in 1h 51m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/26c5279e67424fdb8c27854707466dbc : FAILURE in 2h 36m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/461c9125072c41d8bb6e97b921bdf951 : SUCCESS in 1h 38m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2d7c5689e5e44ef2a6863e5de6624216 : FAILURE in 25m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0a370a6369d84405a57bc7c5c7b9f9bd : SUCCESS in 37m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ce1f1b4253540eda373799bb336b199 : SUCCESS in 1h 14m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/30d9e4a39b6a48f78b00f420e5b1def4 : SUCCESS in 1h 54m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e8f8038bdc29416d81d004e1c694919b : SUCCESS in 43m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/151b1b6f389444f9b34a5603fd14554f : SUCCESS in 1h 27m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3bd1d7ef4d034bef8494e7d4b6ac5d8f : FAILURE in 20m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5622a55c6da84439b7fd6e74a7222cee : SUCCESS in 1h 51m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d295c9627114388b638969dd41eea27 : SUCCESS in 1h 28m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9c4d54e9a544298953ecdc49eb5fdb : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":32},{"id":"56d3d7c13670407bed7347d4288e462e77bbc4c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-04 13:24:34.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/07f754dec4884f14afe7f2e5145dd812 : SUCCESS in 1h 13m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fe3d9286e9ad4d6880d2993051ab1c61 : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":32},{"id":"bb9b1232f483f571fe9cec56d9d3a9b02c65b0aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 15:18:33.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ad6ff611bb40debdc6e49e40ff84a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/445a3f48df7146d1a11d9da4503aa0b4 : FAILURE in 29m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6717a93bde6446aabf1f4ce7bfd05a6 : TIMED_OUT in 1h 01m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3042d474e5140dc8370e9cde60fd9f1 : TIMED_OUT in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"5936a87f81476a0197b88049d1d397df27f51c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-04 17:18:09.000000000","message":"Patch Set 33: Published edit on patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"4cb894f9cdcb69741eb4eef56b2ac45300fad377","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-04 17:37:06.000000000","message":"Patch Set 33:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":33},{"id":"c345cf1e1be5b425322a343818651a9c7eef2d35","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-04 18:07:24.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":33},{"id":"963e881a2fded0b84d3ceb2797030c697b6a38e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 18:10:54.000000000","message":"Patch Set 33:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/846/.","accounts_in_message":[],"_revision_number":33},{"id":"e250d7f4623bc4d7c68ed5b12986d3eaa161c186","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-04 18:14:21.000000000","message":"Patch Set 33:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/970/.","accounts_in_message":[],"_revision_number":33},{"id":"fc2473bcfb4ed33dd2b7606d7458f7998c975d1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-04 18:22:06.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":33},{"id":"3e6a0342958d50999d44cf55cc7ba87b1c6c6d59","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-04 18:47:11.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/606da94cc79c4e3ca0ff9a5d9be44d2a : SUCCESS in 1h 16m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fa1ae619aaf34ccd86e57d28315952f5 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":33},{"id":"52ff082edd849ca912d6a6eb2d8e9351ad91ac55","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-04 19:18:30.000000000","message":"Patch Set 33:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9c8e5e2664534fc998d1d2b0a204666f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17b351d909bb4a00b5b84ba1ad260b8d : NODE_FAILURE Node request 200-0000059161 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f951afe730604531b8dd09c94be459cb : NODE_FAILURE Node request 200-0000059162 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/758ff63d041e45f69eb6c1c92540933c : FAILURE in 49m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9f8ce731af3141dda906cbd31890cc03 : SUCCESS in 44m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/617532aa1f5345539c0365e3a6f5a4b1 : FAILURE in 1h 33m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cddbc4c0a9254db2b9490ca95d9e1772 : SUCCESS in 45m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3d37e2e59e94f54a3db28173e64bcf8 : SUCCESS in 1h 01m 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a9f4f87bf094dcdb098693867cc400f : FAILURE in 1h 03m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/af5ed931c4d34efc8dd47649a47e522f : FAILURE in 1h 59m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0dc42b71b09410788552ac543b4d5b3 : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":33},{"id":"8cd13c7af4185bbe84302af6ee8bdcdb5d177bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 19:56:52.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17e15ae098de40348e3d682b493793cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d044585b81de48b9a7a703900b6ace11 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa21e436152f44568f924efd7ca4ba5a : FAILURE in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6865c0b6b3e7457aa586fc4273f7ff49 : FAILURE in 7m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29b6cc1a22c14217bc148e5bdb8a0d04 : FAILURE in 16m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b08b22095b480a9d0fca041f9d6bf2 : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/c9366525c2494c69a85df208f845df53 : SUCCESS in 53m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb5619e19c31474e93a1b7aaf37adcab : SUCCESS in 32m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5586c97d5fc43b188bb861565b9140c : SUCCESS in 1h 39m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/201204cb2b53491d8d04a063e78a969f : SUCCESS in 1h 01m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3120e61d247424e961872d6dc366cd8 : SUCCESS in 11m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/992563c98d3d4798a2834e23cd02d0e1 : FAILURE in 26m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd059cb3c8eb4b28be27e1f1f879146b : SUCCESS in 7m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0db9b6d89664e0982e984e8c411e13c : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/419561acb1564572aeaffb0c1f260661 : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/683eb2b2533c4e999f6a6710efc3c341 : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/68131cf791da4e0ebcae8ddfda60b0cd : FAILURE in 39m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a7d439738bc49e383ea280ded7a6d7f : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/279aaebe964b4cfe97708827b6617933 : SUCCESS in 58m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d96e0a8b7f2745f1a715343c66b20294 : FAILURE in 2h 26m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20661b19e50c4908ab36879294f0cda0 : SUCCESS in 1h 31m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3a80fd79ef7744378037ea2df3db42c2 : FAILURE in 17m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8577484926e6408a9484255d3b919680 : SUCCESS in 32m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/916d027df0b044118317d800ffb22019 : SUCCESS in 33m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/237fa594d42a482d89d0dfa6ed0cdf4f : FAILURE in 1h 44m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e2060234722f4a7b9cf6324900afbf3e : SUCCESS in 42m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bee8fa8e1e4b4b8c9f835e8a55694b33 : SUCCESS in 1h 13m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5cb4370eaa054b459570dda92cf61b52 : FAILURE in 19m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8878a4d474f442d4baabb300bf8d6da6 : SUCCESS in 1h 31m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/787b59d409fb4d9d92a72426c788fa89 : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f05d06587c5147c0bdb3c4aa260a4b0a : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":33},{"id":"87b16280e2f1a6341f523e1859a0156706670481","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:25:15.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45c5720c09646bd8c698216275088c0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6e800efccd047a38014c2b05affd27d : FAILURE in 13m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a933414b02c74cc981c8a621bdf0eb8d : FAILURE in 12m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63b912b4eec540c48ede2f30d4eddd1e : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"3d9171817dd6fee76f3d2f80213d48b0ebac7734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 06:42:48.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"2e2e36d2b5cb81bcca7b22e1a38bb4a51c3329ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 07:01:15.000000000","message":"Patch Set 34:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/973/.","accounts_in_message":[],"_revision_number":34},{"id":"eb4201fc05671bac184075e7fdb6c8c010b45f69","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 07:01:20.000000000","message":"Patch Set 34:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/849/.","accounts_in_message":[],"_revision_number":34},{"id":"b2a087d14df26873887ce3451c77a1816ce6251f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-05 07:01:56.000000000","message":"Patch Set 34:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"4f6638c1a888ff0fbb9659b730eca00c654f7d84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 07:26:19.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a887aea1893e4994a88c3d0bee3b6474\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b531481d13df4ebc9c2c167aff8e0d0b : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81140645f2cd4da085b3c56c2dc0df7c : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98debc1fefa942668a81d615558a74e9 : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"4f3685fee41efdcc5ca7a6195d0f405fdc8ccf00","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-05 08:11:23.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":34},{"id":"7d15d78b8278791650a1140bdb8973762f05088b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-05 08:39:12.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":34},{"id":"d0c1dc611dc0a249f6dbfd56b0c1fd567113c20a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-05 08:45:30.000000000","message":"Patch Set 34:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4627cb9601664e61a5e02bbf25f7f4de\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b84e4cd118142238166ea70e41bfac7 : NODE_FAILURE Node request 200-0000059302 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7375ad38e05e4f86a19cf318ffae3e00 : NODE_FAILURE Node request 200-0000059303 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a5dec28713848979d0b4f8479e7308a : SUCCESS in 54m 04s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d296eb97c2724fef82f415ca63c613d2 : NODE_FAILURE Node request 200-0000059305 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7874a88722c740d6a6ca0d03ac6d45a1 : FAILURE in 1h 57m 19s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1adae9baecde4255b6a3da063cea2df0 : SUCCESS in 51m 01s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/897c3d97c08f45548b88030577006c85 : FAILURE in 51m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a230931f5e3342129ac2b48ba755f478 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81fd060f43654deba5878e8d37ae106c : FAILURE in 2h 01m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/abdd13c2cd6540e394abfbcab8c909c6 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":34},{"id":"5fa8d49f66b91c87a02f9845f7a65fabef115887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 09:32:45.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443cc4e3301b4f3cbc0f212e9502701d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac1b2c763424354be4955d9108042c0 : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8217fa705d2f49afbe6fbaa59008f36f : SUCCESS in 7m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2f07424d4fb4ec8b6de03e34ced10cb : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0244ba9deee42dca90d43a19cc5ad83 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4681a8cf8904a44a8e58cf2d8a4e09e : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/8cd9e703ca054c27b3d382b232777f57 : SUCCESS in 54m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/121763106a40449e97c10e6faa335959 : SUCCESS in 1h 00m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3ea6e50bd1d43ac845bda5b82f71a64 : SUCCESS in 1h 35m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fee6ded1a60d446999f1c8c229b291b5 : SUCCESS in 1h 05m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf6a67eac424953950faee52bb43e69 : SUCCESS in 18m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17c71653c2dd4a9c87bd20bf258ab036 : SUCCESS in 27m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1220b77dfd14dd99c8ee73c31ff1ab8 : SUCCESS in 7m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d76746ebdd51408489fa25fc5a7451ea : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8b0b579a0c9469097eaaa7c366e5ad8 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ff1e6591ed74037ad027360efc82175 : SUCCESS in 11m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b8490612c6f74792a26fd9872f1da8ce : FAILURE in 36m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8df88ba6a27844259c7c393b2e90a4ef : SUCCESS in 7m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/213524ad62ac4219a4936fd2dfa793b9 : SUCCESS in 2h 00m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8685a541ffb436b8230cf96d43196bd : FAILURE in 2h 45m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a9722a4fa3b4855939004a13392d84d : SUCCESS in 1h 52m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8c59543c7bb3409098f922b13e0fde89 : FAILURE in 12m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3f0e10545eff42749127caadb628edbb : SUCCESS in 16m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a119e26dff2244b48713c4153fbf8ec6 : SUCCESS in 1h 04m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e66d9928cb6741bd897fd04ee6e88412 : SUCCESS in 1h 28m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7922d1c267964385a6e7265257ac09fe : SUCCESS in 23m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f73db73b90674206904d4d240e0f13e9 : SUCCESS in 1h 24m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/28f2d274ca2b46a4b589b6b756a4a7fd : FAILURE in 25m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/738f3c13205748a591542e7a27bc4326 : SUCCESS in 1h 43m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d923a1b673804eceaec1a9c1c0150f42 : SUCCESS in 1h 23m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a233557cfb472a9f9bcb14479d044c : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":34},{"id":"7f21697cfcd7fbfe39df6bdce2c426172212b663","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-05 10:03:41.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf7d2d03afec4795a1e24bbf27b264a2 : SUCCESS in 1h 26m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/49a3b46c2ca443d9b0a114a11deb0bb8 : SUCCESS in 1h 18m 30s","accounts_in_message":[],"_revision_number":34},{"id":"196ede76067770c1404e8eb079d0929d2777f1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-06-05 10:52:29.000000000","message":"Patch Set 35: Patch Set 34 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"38b075652c55c4ff38f1cc4045d53169bcb883fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 11:28:25.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123248244680447a9f5e72243302bb1f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/249152826f054c86975848b8539dbf00 : FAILURE in 28m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6e869ce53f947ea8ba96f9fd9ba2412 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b65311e457e4ee1a46db397a81b406b : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"1e0535a0489169ee64435d043ccc45eee5436fda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-05 11:52:51.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"9cb5e123843939731274c12d0ee2fd5386fb8ce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 12:02:56.000000000","message":"Uploaded patch set 37: New patch set was added with same tree, parent tree, and commit message as Patch Set 36.","accounts_in_message":[],"_revision_number":37},{"id":"a53d0bd41bc9a04f84e338236bfc183b794aa12c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:44:59.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847324010a5c45ae8c0a7238fff864db\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f130e1d6ebf476a8266e01899e01842 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12d488d238184b1b98399a989035e1e9 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30aac7cfc256455cbf323373415ceaa9 : FAILURE in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"d43a9ac9fae7f341cd1af37dac17e172d3de2a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 14:36:08.000000000","message":"Patch Set 37: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4de93441cd214fcc9f3aaf4ee770ba5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1660e174e546f69d6bfb6780536ae5 : FAILURE in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eb8e35f56994775bb976b371e10c351 : SUCCESS in 11m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4275c71bbfd54f8bbb9577b214cc205f : SUCCESS in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/299983fd07d54f19b6b20b7d286f13d5 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f47274edf1a429c9fe62df3160da01d : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/9f91b95cdc1e4d8c9b6b34e4a5b37c61 : SUCCESS in 1h 21m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b794f2a9a0a64070b62ed91a19d2fae4 : SUCCESS in 56m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b19714bff55a462ea47fd63c672ebb93 : SUCCESS in 1h 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8747bd0c49f4d1bbcc4dafc964539b6 : SUCCESS in 1h 06m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3653da16f3fb434ca8a335f62c98fcd9 : SUCCESS in 14m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/10809f02ad184122b2c20b51e5e801f0 : FAILURE in 34m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1e3e7bdfe9434908afea775ce438a6fa : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/252d7528044e41e3ba2d2ad44de98820 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7587346a547d40aa964f12af9fb95304 : SUCCESS in 4m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c9398568517e47e4969ad3371a6154d7 : SUCCESS in 10m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e618a8c6065344b3b299f0034539e6cf : FAILURE in 40m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1295962f4b084bb3be07c1154d57c93a : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d69935ea5b24b70b79c46e114e842ca : SUCCESS in 1h 15m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/747446e2d39847228eb30533509def77 : FAILURE in 2h 31m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f455940e973d4322a3ff3f544315e2c0 : SUCCESS in 2h 01m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/736984a3659e4af9860ebd93a95b7685 : FAILURE in 23m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c5a1b93e60ac41cba8e00049354889b1 : SUCCESS in 34m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a720effc77f429baaabbae314611282 : SUCCESS in 36m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cf7c0ad56b2c4e959d4f0eddce4cd909 : SUCCESS in 1h 57m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6eea35be47444e02acdae40f82c2d585 : SUCCESS in 45m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1488d42456b4f6894de3c50a832ed94 : FAILURE in 1h 07m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/feb0514fd8454e67a28cd541c8a0f1ae : FAILURE in 20m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/789d60af411440efb57128c89a059cec : SUCCESS in 1h 49m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b5a309bca7d4551983f010a5c8dd925 : SUCCESS in 1h 44m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87c707f104b84046a3ad0b248f3126d0 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":37},{"id":"08574853acfa66f9d82e1633b336223eb5015d29","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 15:15:47.000000000","message":"Patch Set 35:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/857/.","accounts_in_message":[],"_revision_number":35},{"id":"6d4417f57188a5bd1f7c08416d18f1f7f296c7e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 15:18:31.000000000","message":"Patch Set 35:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/981/.","accounts_in_message":[],"_revision_number":35},{"id":"6b5487cfa993ca04856a58820a5f23c90a9a6b2a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-05 15:27:06.000000000","message":"Patch Set 35:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":35},{"id":"543dcf1093eff4e1b8fe2128eb44c0dd9a21fb9f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-05 17:59:19.000000000","message":"Patch Set 37:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 48s","accounts_in_message":[],"_revision_number":37},{"id":"6ae18b59d7b67f986c0fd1afd0d1bedee6b81d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 19:11:20.000000000","message":"Patch Set 38: Published edit on patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"739b95cce490f02b03658e246e90f72a7524617a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 19:13:45.000000000","message":"Patch Set 39: Patch Set 38 was rebased","accounts_in_message":[],"_revision_number":39},{"id":"d6525b5722586c3de02e939cc741a02b9d8fa365","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 19:41:14.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3336b50b1c4e4770b47f0daeca7ec067\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b76d5f3e8ad44d66aa5c138690f4bceb : FAILURE in 18m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c3f19c6566843f98c01cc83d7ce6348 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ec8ef47245741328f585d1be0048552 : SUCCESS in 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"4a7830968334805b9db79ba40c9a3f9b0acc5a11","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-05 20:05:28.000000000","message":"Patch Set 39:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/58f93086d42e414ca21709423d4fd494\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e8323a5e1ce484888aa369034b823f5 : NODE_FAILURE Node request 200-0000061753 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0a6cbab6b954235bfabd47e27c157ce : NODE_FAILURE Node request 200-0000061754 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/47d50f6b955241c0a1bb4dd474eed97d : SUCCESS in 45m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/657350a87bb5401da833217a815d15b0 : NODE_FAILURE Node request 200-0000061756 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa6bf4591add46ee90397e81ff5d6b57 : SUCCESS in 41m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fe2b8220a82466a87cf5590af442a02 : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/843e32056ca3436d84813ac0335cc2c8 : NODE_FAILURE Node request 200-0000061759 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b05a535880184b09a37e9e5357a54637 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":39},{"id":"361ed32dbf97e482b8da5c819267858a16a8f6ad","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-05 21:18:07.000000000","message":"Patch Set 39:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-39/FC/ : UNSTABLE in 2m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-39/iSCSI/ : FAILURE in 2h 4m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":39},{"id":"17d58c7c0b3f59f1e5d9374f0ef1478ec9a66f26","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 21:28:40.000000000","message":"Patch Set 36:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/864/.","accounts_in_message":[],"_revision_number":36},{"id":"11695115cfc08c3aa2ebb463d89a200f33371ef7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-05 22:01:20.000000000","message":"Patch Set 36:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/988/.","accounts_in_message":[],"_revision_number":36},{"id":"7910b9a75dfe0fcda26e8efd81dd9199568deaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 22:03:23.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da5064aa043a4e3fb84694876b415bf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e438db3db9e4d13a99f86c5ce8fb2cd : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1841c22ea1d48999c34b40b6e3b3479 : SUCCESS in 11m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d03a61eeb56344d6aa9af73a8bf84f98 : SUCCESS in 8m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb7e306f78d349e2919a7d4d718ed4ae : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1d6dea5b0849548275be71db716751 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/85c22fbee275429c8748f42ae88b0e7d : SUCCESS in 58m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f6389e9bb3644b6b57af1eea3f5ae0b : SUCCESS in 31m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49ce71be2d7e40e9bbd298a234adcf3f : SUCCESS in 1h 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fb1629626ab4388b6465a5a89ac5a26 : SUCCESS in 50m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baadf119e7364c388afe996554fd02d9 : SUCCESS in 21m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92d508cbfd414d2c82e5b938e358db57 : SUCCESS in 26m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/980f56a9b9ec4757a3b25d405ee25669 : SUCCESS in 7m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e654a0651a484fffb6a48d1475bd630f : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7232ea36db0f411da8917290364a2732 : SUCCESS in 6m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/51db5aa0518b4836a61f765b75392201 : SUCCESS in 9m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8579d355891a4a279fbb3d9c0d80fe85 : FAILURE in 29m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b95c515e2b444defadb9259604c2936d : SUCCESS in 8m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ded7ab1e3fd6459d8ab8b7a36dbd4e70 : SUCCESS in 1h 50m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ad79ba8e0aa49a1bac2c81c125eab34 : FAILURE in 2h 45m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f801eacec1a4de9a17e43c2308030bc : SUCCESS in 1h 29m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/882277c417bb44d8a343cb36b2e6daf5 : FAILURE in 24m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e9fe420bce444fc186f45e3066e3f8ab : SUCCESS in 19m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea4740a1b6364070bbea00ea6e735c05 : SUCCESS in 40m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9bb6589718c4c4cac7e23be2411d811 : FAILURE in 1h 41m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2a33374c7ce48c0b5a9b1d9457950d0 : SUCCESS in 31m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06495e8d3d8347b187dad66df45e35da : SUCCESS in 1h 38m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e891fd326ea44db7aa01230cb38b378d : FAILURE in 11m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7bac132b5ca47959b4715157959d9dc : SUCCESS in 1h 46m 00s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ba05e22b02447a48b445a773b2f9d6b : SUCCESS in 1h 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a201a813cf8b474b9edf4fa64ee72073 : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":39},{"id":"31d18915cf3d3d75ad789193cab8499844e352d7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-05 22:34:20.000000000","message":"Patch Set 38:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":38},{"id":"31ea2387da10f290e68e0a42f17c116a940e8de0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-05 22:40:45.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 38m 07s","accounts_in_message":[],"_revision_number":39},{"id":"3bcacdd93456d887b1e4642a6a4492c265ffba73","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-05 23:27:16.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":39},{"id":"bda3cb53174e899cc58ac70b04c331a82b018caf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-06 00:12:00.000000000","message":"Patch Set 39:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":39},{"id":"ef361c1f978116c5ec86f0cba4e574ca024a0cf7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-06 02:25:25.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/08eefa74970b4636afee11e67c27c932 : SUCCESS in 1h 24m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/17ff531ede2a42d7844c091ffc69d300 : FAILURE in 1h 29m 06s","accounts_in_message":[],"_revision_number":39},{"id":"becdc78c564beff8a8cfecc7604d465efa8c1d36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 04:41:48.000000000","message":"Patch Set 37:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/872/.","accounts_in_message":[],"_revision_number":37},{"id":"5c68adf5c54c472ffa431a1ca020d786ab507dbf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 05:30:37.000000000","message":"Patch Set 37:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/996/.","accounts_in_message":[],"_revision_number":37},{"id":"cc992ddd6b853cefad5944d130fb29a08469c761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-06 11:40:30.000000000","message":"Uploaded patch set 40: New patch set was added with same tree, parent tree, and commit message as Patch Set 39.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"d73af898fd01d741cefa7a471f8c30c16f5528bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 12:25:53.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a360d86c28b4cd3b58bed9d94eb0064\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/208095a802c344fbb87aef50634999ab : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/15e6944a15804c5ead5000344084e571 : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb7a65a9e5e0483d8c09f341dea63b4e : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"31e39e25412f32054ff1592c6cbde2380afefcea","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-06 12:41:32.000000000","message":"Patch Set 40:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1453e92f4ddd412fbf2d81a2ba34109b\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ba92e8c52bc40bfb0ac78b95512a11e : NODE_FAILURE Node request 200-0000063080 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f531e7f6db64ea1b627dbb358c62e63 : NODE_FAILURE Node request 200-0000063081 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/06ef453036e548968d52f35c3cfdf498 : SUCCESS in 51m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/251bf46ddcd24d449cc9f97927a2961b : NODE_FAILURE Node request 200-0000063083 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/663472441fc54199866136a445cffe1d : SUCCESS in 50m 08s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64126225ac804c97a47ff6e6fa8435fa : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f47b64c64f3242719a9e287de28a433a : NODE_FAILURE Node request 200-0000063086 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2000ba54572345f391429d7707184e55 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":40},{"id":"65f95711f805f7b12d2fb01c3799246090caa61e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 13:45:36.000000000","message":"Patch Set 38:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1005/.","accounts_in_message":[],"_revision_number":38},{"id":"2e6edba633697c054610515581ce50126361d4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 14:11:36.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279d3666297647fbb6ec3f70379e40bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299a3e6a76a24690ac5f4e0c94237da7 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b24445a308447e2b4a499252e310001 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eed45cf66ec420493f0dc708ceab734 : SUCCESS in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b58b02b743f46058c9e8d3d8b999dce : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbe453af65e46968afdb1be1871e210 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/f07fbacaf2254fb5b07c450ce6d303fa : SUCCESS in 56m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e15cbd6b8954f05a60e2af7fd565a38 : SUCCESS in 54m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec2fa5d22a4b4e2480f2e9c742f80259 : SUCCESS in 1h 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0a3665785544084b479ae3270734c9a : SUCCESS in 1h 01m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3da52dc9cad743c0b593ad5c14975553 : SUCCESS in 17m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/172cfdde867b43d0b9703a8536c1ec3a : SUCCESS in 24m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/868558afe68041798a71376090d0cf55 : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c868c9ca1ed74050a8625ded1a88cdd5 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6fef82af7544de5a9fd44a853d1e5b2 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21c6c54d100c4e2aad3064f7e778fa56 : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2adf73026d14088ac1d036296637a41 : FAILURE in 31m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/137783877fbb47eea3bddcf617c39d2c : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4291ba8b49c0426dbcbd16b3e4bddfee : SUCCESS in 1h 32m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8cb96d2787fc42ab9eac53154e691ad3 : FAILURE in 2h 15m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5aa18b9783c44021ba59eb49110795e3 : SUCCESS in 2h 19m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1b41fbaeb7924f938d7201518c5140e2 : FAILURE in 23m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c5bef7737ab4b64ae5ea4740963bdff : SUCCESS in 36m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8244cef35b784718a70ab1e2352b1141 : SUCCESS in 1h 07m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9bc12a3ede5b4fa081bdb0e8e5adab20 : SUCCESS in 1h 25m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d6e2584cace445da23fb8ca148961d3 : SUCCESS in 35m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0ee8707210c3498994cad349cd07d5cd : SUCCESS in 1h 08m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/600cb1f3038143ec94952a7a4d11ceca : FAILURE in 23m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1fcfc618fdb4843b30d60e1f1e5b170 : SUCCESS in 1h 12m 21s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1b5c31047ca464ea69434b83d75b203 : SUCCESS in 1h 45m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a343be12390f473993e1035b7b37ca37 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":40},{"id":"03a678b9fc1025b6d6113128ecfd2fd99cd647a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 14:24:38.000000000","message":"Patch Set 38:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/881/.","accounts_in_message":[],"_revision_number":38},{"id":"85ddba37b89cf930f63e5bec66925a7b4cddb14b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-06 14:35:01.000000000","message":"Patch Set 40:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-40/FC/ : UNSTABLE in 4m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-40/iSCSI/ : FAILURE in 2h 54m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":40},{"id":"2087db37e57ec3920107fc4bab52c3f76a2d17ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-06 14:51:30.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 37m 47s","accounts_in_message":[],"_revision_number":40},{"id":"a2399b55c7de436c971ef57d62214a00dad23eba","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-06 15:24:09.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 17s","accounts_in_message":[],"_revision_number":40},{"id":"92f7d14f9d5a34d1d98fd4a4cc5422ace50ab2d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 15:39:26.000000000","message":"Patch Set 39:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1007/.","accounts_in_message":[],"_revision_number":39},{"id":"4eec71617cd36ab37886c61cd369fd75109411dc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 16:10:43.000000000","message":"Patch Set 39:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/883/.","accounts_in_message":[],"_revision_number":39},{"id":"7615946552a22752cfb9abf057ca8ec2d0e1c363","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-06 18:21:45.000000000","message":"Patch Set 40:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"5838cb100ac3491cd50c429fd4de21e03c69f732","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-06 18:31:37.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/417587d6a8df4f63ad04f8e8e8148c47 : SUCCESS in 1h 23m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/205d7f77a874469895971653ae237e54 : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":40},{"id":"722fc0427d6caa9b8cd34d3dd55b2dbc273ac811","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-07 04:37:55.000000000","message":"Patch Set 40:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1022/.","accounts_in_message":[],"_revision_number":40},{"id":"13099b0ac3fc6143e88e3a674773723041ef784d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-07 05:29:45.000000000","message":"Patch Set 40:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/898/.","accounts_in_message":[],"_revision_number":40},{"id":"4b3249dcf1a41d592d6f4fc06f674d67a62465f7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-09 07:15:18.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/40/20250609113853/NEC-V-ISCSI : SUCCESS in 4h 20m 02s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/40/20250609113853/NEC-V-FC : SUCCESS in 4h 36m 22s","accounts_in_message":[],"_revision_number":40},{"id":"59c73923335f88941b3569de3ef4c98d1e5d21fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-06-09 10:53:50.000000000","message":"Uploaded patch set 41: New patch set was added with same tree, parent tree, and commit message as Patch Set 40.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"8e8fa06e60f9e1c2babd8375d45fb0a83fac8054","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:55:55.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62087342df5a4abe8b240e2abc287254\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f6c2f5a21b8431d93e8fac712470d8a : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49ae123b41824d09aa86931c96375ba7 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c76af759e7994f01b3daa9bc5f0e3dfa : SUCCESS in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"494fa379982806db3023bd67955fb90a7058dc0e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 12:32:20.000000000","message":"Patch Set 41:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5cd4e6b7f71c4e718287bcca2d47c5d7\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d206c5742be4c0f9c01c3bee6db5535 : NODE_FAILURE Node request 200-0000064901 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1bf22f135a584c06a85be5d39ecdb17d : NODE_FAILURE Node request 200-0000064902 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ee210e63c4f4e4192cd2c4617029e27 : FAILURE in 43m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76333fd5d4e4495389595330531f0d0c : NODE_FAILURE Node request 200-0000064904 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/51c407345ea4418f816d6137194328e5 : SUCCESS in 47m 36s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84b7292015f046e383f1308d157670db : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/759784d9f85f4282a88b4d26f22e6835 : NODE_FAILURE Node request 200-0000064907 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80fdc6f6a8eb4b33985d692758daabc2 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":41},{"id":"1640db79b5586d1a720b102d372f974bba0c2908","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 13:25:25.000000000","message":"Patch Set 41:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-41/FC/ : UNSTABLE in 6m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-41/iSCSI/ : FAILURE in 2h 31m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":41},{"id":"6b34be722e8f176258e53495895c81758afa284a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:42:24.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a49c886ad8dd42fba9bea07fb1421ec2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f390c7cad748e9b9110297481b4c80 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddb585bef94b4e62a7c8b4af7206b8d5 : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da21d9085e3b48ff85af50b717f59968 : SUCCESS in 10m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a4704bfab244d179f30ab62540d8982 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf384b87745400bad55a4248bbf1650 : SUCCESS in 15m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ac9580273b26422088245444a50b72e9 : SUCCESS in 1h 00m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0cb1bb3e1f954f8e9417e576818d5367 : SUCCESS in 34m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bab13939eef44029aa61d212fac3d7e4 : SUCCESS in 1h 20m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b704a562750a4295b8a9fba773e7faa3 : SUCCESS in 1h 03m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522b342ee5bf4e7b8bdc63647e061880 : SUCCESS in 27m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e592f7a2aed40c486f539086a488c32 : FAILURE in 28m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1aad51e3d1dc4c30a9eb54ea9bdf7a4c : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b7ee5822c08c4f72a95f0fcc547f4861 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38fabcbf512f49c0b02446ec58451e1e : SUCCESS in 7m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b91bca5503794698861af228ddbbfc27 : SUCCESS in 9m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2f98ca5db72444b1976b5a38758be2df : FAILURE in 33m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9cb9a1c90784d3e92401ec0fd2d4756 : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5db7ee01f7c04a5084d66adbce4ea01e : SUCCESS in 1h 59m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf244954b50c4db09e3aa1be71d23aab : FAILURE in 2h 36m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c232e63298c42cfa2ea3ceb1faa652b : SUCCESS in 2h 20m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f25927cf2f974f89b38c86af16ffa968 : FAILURE in 23m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/65cde6fb91e1435886f89f650c287675 : SUCCESS in 29m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8add8ee87874532b0375ab5217dc399 : SUCCESS in 59m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1354bf9e50c14dc6b72d017843642814 : SUCCESS in 1h 38m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/08cb023226bd48a3abb55fe25bcd2a6c : SUCCESS in 38m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ecb38fdfa83f4a1f93417a842cbcfe88 : SUCCESS in 1h 28m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e76b2dcce6964b4d9e33bf1ad3882d10 : FAILURE in 25m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0ddd44e01ea40b0905919cc484d7d4a : SUCCESS in 1h 38m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781547f5809646a5b3dd9165742f8b33 : SUCCESS in 1h 25m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e336d87dcaf4dd6b1acd1bd8b44351e : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":41},{"id":"1dca0d70a295ab80096503614ecbe48713d172c1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 14:19:34.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 55s","accounts_in_message":[],"_revision_number":41},{"id":"e2f2c26a7be51b9f144fb43b7e8cfa44ebf06b84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-09 14:35:39.000000000","message":"Uploaded patch set 42: New patch set was added with same tree, parent tree, and commit message as Patch Set 41.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"28c4d02b99297ba278e539a758d497b3d9486e5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-09 15:05:24.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"d830efe08f27a30252cfb689c302bd41f5a602b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 15:32:48.000000000","message":"Patch Set 41:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/911/.","accounts_in_message":[],"_revision_number":41},{"id":"92aef118c71c5bd6292590aca8b5e6d12260f7f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:38:51.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c58912ee95a043688b6616256cfc3132\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6c470b2a32045f1847b84b80a459e51 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96031ac3b65d476b80d83a32806d4ce9 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9d8a0e02a964a40a8384ba967454b29 : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"292c1a94180d4383287fea53af4b01e7af06d07b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 15:48:57.000000000","message":"Patch Set 41:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1035/.","accounts_in_message":[],"_revision_number":41},{"id":"c7f1f0faa3f0a4a3b402f0295d4131b63f227a72","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 16:06:42.000000000","message":"Patch Set 43:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/74374b7b77a740f7a23c9cf88410e02e\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6744e3a5e314f478d8ac780ec0fa4d6 : NODE_FAILURE Node request 200-0000065537 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07e291ca3b3b48a6b7c871ca3e099e48 : NODE_FAILURE Node request 200-0000065538 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d30a7e48e6d43d0a9ec1d530e1da99a : SUCCESS in 56m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65852809c75e4244895e1a720e8b5729 : SUCCESS in 45m 59s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2fe4b32cb9645ce99eb4477a8fc9f5e : SUCCESS in 54m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84c23a283c034aec811a3ec0f5f18d16 : FAILURE in 52m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d53593a4da04511a4acda03f8063321 : SUCCESS in 1h 00m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8dca5e7402a4a79baf097e9792c2244 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":43},{"id":"6ec17e93aee864f1162bb2b7ca62cd542d41a9b8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 16:38:18.000000000","message":"Patch Set 42:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-42/FC/ : UNSTABLE in 4m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-42/iSCSI/ : FAILURE in 2h 2m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":42},{"id":"755758e612859b9870515654f1f1775fff640845","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 16:42:56.000000000","message":"Patch Set 43:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-43/FC/ : UNSTABLE in 2m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-43/iSCSI/ : FAILURE in 1h 37m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":43},{"id":"42e63f68b37a3fae2448c235f6f6fba9083c992f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:00:02.000000000","message":"Patch Set 43: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9a303c7ae9e4d13a596f0b608a5f926\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6241a1b5e40345ef8182a31032870cc3 : FAILURE in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ac0d46ccc643ce82ee07b2872b20fc : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64f10281d5df4585a52db3b9811ac97b : SUCCESS in 12m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef0721133e9c439a812e3b0a3848a30b : SUCCESS in 17m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c36ea1d786438ba001364f9d4f43af : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e5d699500ced455fb1097baf06da1e7e : SUCCESS in 30m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d7d072df3bd48af8b86dc83c05317ee : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba8d7db3540646c487e59dff969745c0 : SUCCESS in 1h 40m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33c21079cba4d0b937c344ff18008e8 : SUCCESS in 58m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f27136826b8d48f8837507f8534fc010 : SUCCESS in 16m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c2729ea90244893a83ea8544ac36138 : SUCCESS in 25m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9db73f787752457d9ab3346495fbbee4 : SUCCESS in 3m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e8e9d1e99a44712b55a4dfbf4635382 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2e5d97043c424b12b6ca9c59b80c24a5 : SUCCESS in 4m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e3401e1996b494c9aee4490e440ea15 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc4965bfda9c4b62bb6bf0e55c51aa17 : FAILURE in 18m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40c6d7b12ed0433ea7a1011bd40503c2 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5eacbc0255cb456186ca3c46fde53b3c : SUCCESS in 1h 52m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab1114fb8f4747409b1b0397458cbdaa : SUCCESS in 2h 49m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/03faacc403964359a2aa16e1711e3c10 : SUCCESS in 1h 46m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/90d1fd8048374071be5670e8b3b34cd9 : FAILURE in 24m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6998b6eb3e844af596b262e28d9f90f1 : SUCCESS in 21m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06df2e5fb5e3408da60ec20379f457c9 : SUCCESS in 1h 11m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7e87fa3c21f438cb0828a9175680e31 : SUCCESS in 1h 47m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/397423c3753547e1b02b493b5a9187e1 : SUCCESS in 28m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d97437b0df8f43b9bf3994548e9be763 : SUCCESS in 1h 41m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/87d13986e9f94d47a98c0e4c589041ae : FAILURE in 14m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/534509834be0490c86fe84c8517d12fd : SUCCESS in 2h 05m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa378cddf7c441dda5f7b494d7d53fa5 : SUCCESS in 1h 32m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbc0a4d391c745358e1cd514917156fc : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":43},{"id":"51f2ead7dabf9bb639acb0b4bc73a07f51d63523","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 18:43:43.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":43},{"id":"ff45009d95999b81524fdcc1452ca6ea4a9c62d1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-09 19:38:45.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 13s","accounts_in_message":[],"_revision_number":43},{"id":"48b404f6889f71b3a6b8e28734b2cb6e8c441966","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-09 21:49:23.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a9be161c4da140fdaf0193978f35650e : SUCCESS in 1h 16m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cd956d1f81b04451ade2a8955f9174d3 : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":43},{"id":"5792e6411f889b59d4b70962e94374fa6d67fb24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 23:36:17.000000000","message":"Patch Set 42:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1044/.","accounts_in_message":[],"_revision_number":42},{"id":"6828d922a3ef98f7eca95213ea488a0ec899bd6c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-09 23:47:05.000000000","message":"Patch Set 42:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"809f773026280c9f1d25a7c7daadf158437d50d3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-09 23:50:27.000000000","message":"Patch Set 43:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"fe2ce58e4cbb2c592a6e193f70a766edef72750b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 23:50:42.000000000","message":"Patch Set 42:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/920/.","accounts_in_message":[],"_revision_number":42},{"id":"d5f0017274f36fae5f644d027a341170f68b6f2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 00:33:00.000000000","message":"Patch Set 43:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1045/.","accounts_in_message":[],"_revision_number":43},{"id":"6aed44edf011ff8c6c0bc56f7e1e99add32a21db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 00:43:05.000000000","message":"Patch Set 43:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/921/.","accounts_in_message":[],"_revision_number":43},{"id":"7b7aa04826e0a5a2d760198370944ecf59b59b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-10 06:46:09.000000000","message":"Patch Set 44: Published edit on patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"7598bcd6a4980809e126fa7895a7491701b4c395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 07:44:04.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56571bfb71ef48bc800c4fa99c764f29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5690c387f6184551942261bc1835eafc : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e62a97c9151483cad0ca74dae41da7a : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b2cd271299946be92e96467663fb565 : SUCCESS in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"a53d96ca2b27f702c9cb7219747ed0631165171b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-10 07:52:57.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":44},{"id":"d1f71e0307a063f216023b74df70c807a9bfd90e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 08:19:06.000000000","message":"Patch Set 44:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/932/.","accounts_in_message":[],"_revision_number":44},{"id":"507f1caa092bdc10eb18088e3fd70281c3bf3dae","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-10 08:22:26.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 20m 05s","accounts_in_message":[],"_revision_number":44},{"id":"ca4a03fe038d874b042fe3dc28711399f45795f6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 08:23:28.000000000","message":"Patch Set 44:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":44},{"id":"25217c5068c740adc696a4a394d0a280740c5133","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 08:27:35.000000000","message":"Patch Set 44:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1056/.","accounts_in_message":[],"_revision_number":44},{"id":"73248ad5d50ca0bbfbce8ab5e40f4e51e36705c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-10 08:28:20.000000000","message":"Patch Set 44:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f4f84bf3fcbc4682911dccbe715a8507 : SUCCESS in 1h 28m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/47b4f5be8ad24beb905cb40ffaa58cde : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":44},{"id":"7dbd260146c1311c02e3c7478a561b9c0c199fdd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-10 08:34:50.000000000","message":"Patch Set 44:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ea53cf29fb60468d97b8beccb16bc186\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f8e23380ef9472e872f1367cd148d68 : NODE_FAILURE Node request 200-0000065764 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ea4350a80b84009aa8f0b5525c77c39 : NODE_FAILURE Node request 200-0000065765 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a707666edcc74674a378e700002c7a68 : SUCCESS in 55m 07s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa0043a250514e05a515cc9f18b1c7ba : FAILURE in 1h 47m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/354ddeb0bb7d4ad59983d81bd613c298 : SUCCESS in 47m 21s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fecb3de739984ea89515fb1a5f06de8a : SUCCESS in 50m 57s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6449c843172049e9b456fdc30930546d : SUCCESS in 1h 02m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddd21a84886940879686ae13e14e2aca : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":44},{"id":"d4f0d08afbf1ca414d72fb5ad3791c2ba9e57598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 09:20:28.000000000","message":"Patch Set 44: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdf27f22459b43eb8ad5946030d2d6eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f03eb0bfd3f948249161dc91a31b6068 : FAILURE in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ac21da531cc483ca7fa2864f1f37387 : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf4dffc8543b4ef4bef7ab65c5789796 : SUCCESS in 10m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fb094bebdf44c5180d643d41f3e3e01 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78d5a2b4ab34878a7a174117fc275ea : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/cd3e9a952c4c497ba9066a878e997368 : SUCCESS in 53m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64ddf1bcbffc4b54bc64f4ca43b117ac : SUCCESS in 53m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed992dd807a64ab4903c5c80625f5fb2 : SUCCESS in 1h 42m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd6814a569f248c7a89ecc0a7b5c7d42 : SUCCESS in 56m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfe1b64e36de4d83a1f5393a646e2efa : SUCCESS in 19m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/83d7b1512219455d8ed29ca4a1917b5e : SUCCESS in 23m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8547d47ba2c04c45a1eba6ee868bdcbd : SUCCESS in 7m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e3f430825714fc8931f6b208cd1ae17 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa31c240de754a1abef6233e37ca385c : SUCCESS in 3m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3da1244164df45f68fd9bf07af3f9ecd : SUCCESS in 9m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b50b643bc6744f84a14bc3a5fe43d6ed : FAILURE in 33m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/862bfccfa4ed4b40a563322e3b391862 : SUCCESS in 8m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7fdfffa4b214f23b36aca80ed9904d9 : SUCCESS in 1h 35m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/15d8358595b14c10bd658fbe9e743544 : FAILURE in 2h 31m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6c00698eee544eead646609e80c6e8a : SUCCESS in 1h 39m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/310555d07c794e2dad05c74b93b59830 : FAILURE in 22m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a05bdf40c4e44580898b958df4c5812e : SUCCESS in 27m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfc6b2737480444e92dc9e24d2ea6d7c : SUCCESS in 1h 07m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94b249bd8de94d049766d6b91cce4520 : SUCCESS in 1h 35m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b59e48b9f2f456fbae8382f4c22e503 : SUCCESS in 36m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e6d4e5f6f974950b9950180a55a5e68 : POST_FAILURE in 2h 01m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/346ae1c1a7e24307bc91143cb3b650fe : FAILURE in 23m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/388f04382c3b4bc4984676f2c6129d11 : SUCCESS in 52m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/596e8164e4fc4b63963c5556c34acecb : SUCCESS in 1h 49m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c7cfb5d4db644b0984e781d15b84f33 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":44},{"id":"32054459889a11ca8d5e8d36f82f2b4278b01d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-10 09:59:24.000000000","message":"Patch Set 45: Published edit on patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"4ca8ea9b3e486e9696a4cc5ed377a2975345f62d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-10 10:59:35.000000000","message":"Patch Set 45:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/077d8e79ae864dcfa966b19c0dd8305e\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07f19e29d900403f84f5c9197ee31172 : NODE_FAILURE Node request 200-0000065815 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02ce871bf0b04e458e19b51b042ccedb : NODE_FAILURE Node request 200-0000065816 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/002cb216f14c46df856381284a4ac53e : SUCCESS in 46m 11s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/249247dabb4144c3b28aed0efe4019b6 : SUCCESS in 46m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e098733aab1420db9734b21487f5a56 : SUCCESS in 48m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9e61c12d0ae4f2e8542a6c742bc503e : SUCCESS in 46m 12s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7377db906d440558bb49ef11793a983 : FAILURE in 59m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/856dda4cff6e41a784f5021c30d46659 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":45},{"id":"01f1910e9291703174adec566d2065180174c3ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-10 11:15:26.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":45},{"id":"b068aeafc2def10f657e03cf10f30a001cbadc11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:16:38.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1111d88d56b44fbbbfc4a8ca75ab7f2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a5f154780fb4fdfbdd43ae783882b6d : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fc247aa03a74b9ebfa55673cdafb80a : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b821c3dea04417ab3cb9f63114b0e92 : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"e28f8299f2e77c03847ee49f7f4f5a761578780a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 12:01:35.000000000","message":"Patch Set 45:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/936/.","accounts_in_message":[],"_revision_number":45},{"id":"158ed1eb98261018fc19488095c4b73360c0d4a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 12:09:01.000000000","message":"Patch Set 45:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1060/.","accounts_in_message":[],"_revision_number":45},{"id":"948823efb1a3e126ad64aa5dae8f179cb3a78411","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-10 12:52:44.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5c95068024c74029bab74729e54b7b27 : SUCCESS in 1h 28m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a08f8103943c47c98aa44780e32f31d5 : FAILURE in 1h 28m 35s","accounts_in_message":[],"_revision_number":45},{"id":"ee2e55a81afbe50a666f9d129b1ae058599dfded","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-10 12:56:28.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 1h 18m 50s","accounts_in_message":[],"_revision_number":45},{"id":"29ab0ce9fb78301512066a91ae4b0e5ecb5bba04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 13:02:11.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bb2983ef454995baa2d724fcd2221d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ffad92788249b1a0f8b399cf3c478e : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b6983ed0a6d4e5f9cc92eb95856e772 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f607f905d54904b9a933cb361f313c : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58305badc5794355bf5e811df1431a01 : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cf647197cd4095b72d0ccb6349cd7f : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/38b8d17b1cd94b488654fd98de57ab27 : SUCCESS in 28m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/760adcb611db4fd5a91f6af6c8ffe157 : SUCCESS in 57m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5710ec3c4e744888c06fbf9fb5ae9b2 : SUCCESS in 1h 33m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3200672d34d8409088bd91f0cf006a06 : SUCCESS in 54m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aad2cb424c64e6caafa80cb5c7609d4 : SUCCESS in 19m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1165e0c0347341ae85021eb9e349507f : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f6f181f516c407aaedd4daa5a3c6479 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0be139887b244a439721c9642fe565e5 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07251731774c49829d3be722d6381236 : SUCCESS in 5m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7dbadbe0d1734a098a758371d78347a0 : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf3bb47520814442851c8691e075eac0 : FAILURE in 32m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93a28e36830e4bd68bb89d8dcc6bc2d2 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7512750970d344d388e59aabcc43551a : SUCCESS in 1h 45m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d8cbe0708679430c9793838ecaf33630 : SUCCESS in 2h 49m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/142f9d23bf8f41f88c7c585ff16e8fa9 : SUCCESS in 1h 41m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ba38797d1e0643e5baa40a14d4df09ec : FAILURE in 19m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b5c4dde6661c40e2af7b63d6d8aa31df : SUCCESS in 39m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5cd69060b5214b75883edd67860dff7d : SUCCESS in 1h 09m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42cc2a2391c34a72bdaed512410419b3 : SUCCESS in 1h 34m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4397ec7437d641c6bb76327dea523316 : SUCCESS in 43m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eb5d455c85dc4e31865a61e0a23382a4 : SUCCESS in 1h 22m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aa2932ae1072476ab7dddc1f1e8f92da : FAILURE in 18m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9276ea88a174f8da35c841205bacc70 : SUCCESS in 1h 33m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4b3b2c1cfa343c3b6688c52b62d7092 : SUCCESS in 1h 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77afeef15c294ae5a6752a0d1daa62b0 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":45},{"id":"f415f61975cd3d3c93f87ec35b2f80f9354d12a9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 13:29:08.000000000","message":"Patch Set 45:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":45},{"id":"8bd6b38c4ad4a78ac89be410535ff0ac394b5f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-10 16:11:05.000000000","message":"Patch Set 46: Patch Set 45 was rebased on behalf of \u003cGERRIT_ACCOUNT_36538\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}],"_revision_number":46},{"id":"6add1e51e7f3a5725773ec29a69284cfbf5b7051","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 16:52:58.000000000","message":"Patch Set 46:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/942/.","accounts_in_message":[],"_revision_number":46},{"id":"3b1cc9c43989583376e2925e08179174a4ab2a45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 16:54:52.000000000","message":"Patch Set 46:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1066/.","accounts_in_message":[],"_revision_number":46},{"id":"124750b3e273fa7d61626f9966c5a73c6515d615","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-10 17:11:34.000000000","message":"Patch Set 46:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a562e2c326014539b0b7385562446cda\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/321d316e37de41adb4352fb9161ec02d : NODE_FAILURE Node request 200-0000065882 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d505f9238cb14b8d88d75b9e8aff0f38 : NODE_FAILURE Node request 200-0000065883 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9f29f5d8c7447598b5f16c53875a92f : SUCCESS in 48m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/05781b8717c34a4b8446d90aab9c4ce9 : FAILURE in 54m 49s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dcf8eb7d25dc4c21bb68727ed392b4a2 : SUCCESS in 44m 26s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0610ffe6f6ce4d6996a06dd300ed3bc5 : NODE_FAILURE Node request 200-0000066143 failed in 0s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf426b8d67094e058b7a5eb3d3cb56af : NODE_FAILURE Node request 200-0000065888 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5bc9f7067c884c9e9b9742edf03ded22 : RETRY_LIMIT in 38s","accounts_in_message":[],"_revision_number":46},{"id":"ef2a8ad2ef4de33332872570fb208689c00b7d92","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-10 17:50:33.000000000","message":"Patch Set 46:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-46/FC/ : UNSTABLE in 5m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-46/iSCSI/ : FAILURE in 1h 39m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":46},{"id":"2c8f8e9172461d7e3f3364f9fc2fe07e5586895f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 18:08:26.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/706fe8ef3d96411f9b188f94ce9f9383\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf6b52744ee4aecbc7524e7cbd3123e : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f886cae4ac754d9b94ad1b7d3dd30d59 : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cabd9b9c70c344aca6f96bed92568213 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4db0f57535cd40848818f788e93cd873 : SUCCESS in 15m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142852b7bdd74a648d7fe63eb068d3cd : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2bcc1725bc69425492674b44e2454e14 : SUCCESS in 54m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1a9268fc5a04064bb1a81fa2b65ac2c : SUCCESS in 1h 01m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a168325552724b188f1b9fcec91ca9cc : SUCCESS in 1h 39m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/698c5c2f49764d00bf0ca2bebfe53ba4 : SUCCESS in 38m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f06d45e0b646b7a0bd3b79ab599a42 : SUCCESS in 18m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab98bf480e8346c7a1626a375e44eaef : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2bfc6f80858447a098e68bdef0be488c : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a16f695546a4740a3b74cc60672d886 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/061a4372c1f84b4e847b59bab60788e2 : SUCCESS in 5m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/18b096ed427a42b687a21626502dcd14 : SUCCESS in 11m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1896f018c664152a9ceae3706875de2 : FAILURE in 36m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dca46c0baed84852984fa69cd9681913 : SUCCESS in 7m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5e7204eccfd1403f877b006f84cac485 : SUCCESS in 1h 51m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/951c7578aa294faca32a304f08f6112c : SUCCESS in 1h 34m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f90eb48c20e545af972ad7fe87286c87 : SUCCESS in 1h 42m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5959e724f8d7409291b3ea6341b51007 : FAILURE in 19m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3aaf27df06bf4e5ab1e111a40a073f2c : SUCCESS in 30m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae8a0df3bc9e44ea9af960fa80a2bedf : SUCCESS in 34m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e830b8fc9fd948e7ac3b4c61cd30f5c5 : SUCCESS in 1h 28m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b72441e7a5e243a39dd0663872a03bbe : SUCCESS in 39m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/245035696fb042c5a29552d23806776c : SUCCESS in 1h 31m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/650e25e214ea4983bc18629a241e9be6 : FAILURE in 19m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afffd48cbe85447ea15a1634f6c16140 : SUCCESS in 1h 36m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5cd0bc96e244ccca2e1c71d61be7089 : SUCCESS in 1h 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b30feb88e9c4213bb3a8b30ed2ad5b3 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":46},{"id":"1f65c847d4972aeeb37d9370c65c6c4b47492511","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-10 19:18:42.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 37m 33s","accounts_in_message":[],"_revision_number":46},{"id":"fe05fca48a94aba2e8f40c3b545732ed37c50443","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 19:44:20.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4442c4c6a91e4302b9dd5551c600be46\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/437aaf84f83a4b84a552f1bc78b5a00c : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d1d1b36a6cd343caa328135dc4881c88 : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e4ff90131584c39bc9ee33721ec0f2e : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"d384e8aae9195722fa2f436bb1094cbe1065a56e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-10 20:13:19.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 29s","accounts_in_message":[],"_revision_number":46},{"id":"f0d88b75263a365b41a94dd06b0556aa2a2cca28","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-10 21:52:39.000000000","message":"Patch Set 46:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f261589bdbf44b3dbfd164a9f0e46fdc : SUCCESS in 1h 19m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/632137b14c994b39877d25df863bb8cb : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":46},{"id":"d16dc584f62d3bf571ebab9612c4bb6c3af20dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-12 08:53:15.000000000","message":"Uploaded patch set 47: New patch set was added with same tree, parent tree, and commit message as Patch Set 46.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"65121688d565dbfbf9a5faff11e411879ab88a94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 09:34:25.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b7f468a8ead4d6483364fcc86ff78dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89ffd91f9fc84425b7afc0dfa555eec8 : SUCCESS in 25m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d40d142a65bc4295b3dc34c7bbea9bcb : FAILURE in 31m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c181d149a374562941a3b26162cf404 : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"04ce69c01d18f64d37acccf865346a0e8ec44b35","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-12 10:42:51.000000000","message":"Patch Set 47:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/11bb8c1242b04f969df29f82fe677d89\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d80c42432ac4b69836891bda6478441 : NODE_FAILURE Node request 200-0000067084 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b5451c72a90489fb941f30acd6bd8b3 : NODE_FAILURE Node request 200-0000067085 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3e2104761b0748da8f151bafd45fd956 : SUCCESS in 47m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b21e293a886a4c5d854da7b738cb01cd : SUCCESS in 51m 17s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/525e3d1e7141494d8e29aa8010fdc612 : SUCCESS in 45m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/591257b5b2554b378a31e293e23ac81f : RETRY_LIMIT in 31s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca6c4e672b934f848abf4f55b7cc4ebf : NODE_FAILURE Node request 200-0000067090 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ea5a922cdec4708a628b2fe63856793 : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":47},{"id":"aef8c6277bc4f2b73913de3950e6ee41d05f163d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 11:14:32.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12b224fa5d014d0a91d9aeb572d60638\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d39a6afd7946fa808efe722dbe94b4 : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/015a3f897df845a58c313bb084fcbda5 : SUCCESS in 15m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38f620b98b8245378b5d74b93802d6f1 : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c83dc2370b904d3ea777f2d26cf78ae1 : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276e79d4769a4362a818d7d4806c6611 : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d7d0965163794898985b0efd114fb7c4 : SUCCESS in 55m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2a7da7f89d1418c9e5ae2b61d635959 : SUCCESS in 1h 01m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62d8625b1a6c40c68223c0463dfb0d2d : SUCCESS in 1h 31m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9bfe776f527437c97c4548e80e38c5f : SUCCESS in 59m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995264aaa89b444880218e8988565f9f : SUCCESS in 19m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ec753b8428349d4bdda4418fbc3ffd1 : SUCCESS in 26m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/332869579a484cda98edd2570e13861d : SUCCESS in 8m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3351f93ca66a42e7b52df19b89cb4b9b : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fea79210eb52465aa1c33ee9eed9d9ca : SUCCESS in 7m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b6c5b3b198574d32a89bafcb2a1b9b3b : SUCCESS in 9m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1e1e42d685e4552ab1fda05ecff1544 : FAILURE in 37m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23769d4b7ea9454ab6aec151cbd9b51f : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/774c3b0c9a264ebc9085182848920d1c : SUCCESS in 2h 05m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/859e943721974c9ebb626748c5944646 : FAILURE in 1h 59m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea13626ab7db4e8d9e8c255c5bd48260 : SUCCESS in 2h 03m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/11866aa809cf40dda4a68bcea06f01cf : FAILURE in 21m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/000dc8ae5ec040608859431303c5660f : SUCCESS in 35m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/524723df7859445493a4e760629f2e5e : SUCCESS in 1h 05m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/893e143c497a420d8bbca34b94311813 : SUCCESS in 1h 08m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7c8ffe140b7416e92fccaa8274c7ce4 : SUCCESS in 40m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b566d7df8334a3890d7bd1b3763e933 : POST_FAILURE in 2h 12m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0e315540a88f4d60b8ee63c963f61d04 : FAILURE in 18m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d9bb01c09fb43fbb69ffbc29e1e2c5e : SUCCESS in 1h 39m 36s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67803084112a4796ae829f02a54bc867 : SUCCESS in 1h 34m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba1de2d2c69140aab0e781f5dee67f04 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":47},{"id":"2e42003083912d7c1ce30c9b9d2ce317b37c2b97","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-12 11:18:08.000000000","message":"Patch Set 47:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-47/FC/ : UNSTABLE in 10m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-47/iSCSI/ : FAILURE in 2h 24m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":47},{"id":"7d634d292326a4c4ba52e9270e8601f594274378","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-12 12:10:27.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 38m 18s","accounts_in_message":[],"_revision_number":47},{"id":"7db1527abfd93416b13ae2743763f4fec703506c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-12 13:05:04.000000000","message":"Patch Set 47:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":47},{"id":"0f7cff0e193a4174fde9cd822fbb10de1e93574b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-12 14:08:17.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 34m 50s","accounts_in_message":[],"_revision_number":47},{"id":"d948bf96447e0bf8e3f0c2ace87a4874dc79c84b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 14:33:33.000000000","message":"Patch Set 47:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/967/.","accounts_in_message":[],"_revision_number":47},{"id":"a5dfdae762b43fe322a88978a02d840e68f988bc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 14:42:22.000000000","message":"Patch Set 47:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1091/.","accounts_in_message":[],"_revision_number":47},{"id":"3e1fc56175429e5a0f4ade9489328a5001948f17","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-12 17:13:39.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d9ea6d50e2954b41a4941f389e1c5e9a : SUCCESS in 1h 20m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f053d94fd5094ff3b5a0bb2b53a142ee : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":47},{"id":"ff30c8f3a6b6d0ba9aa3d78485ee80b302b7c3a8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-17 03:36:05.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/47/20250617081039/NEC-V-ISCSI : SUCCESS in 4h 18m 07s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/47/20250617081039/NEC-V-FC : SUCCESS in 4h 25m 23s","accounts_in_message":[],"_revision_number":47},{"id":"83dd1c2ba91532d7971c477a81dafb6a5f190dee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-17 18:19:58.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/47/20250617210438/NEC-ISCSI : SUCCESS in 5h 51m 48s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/47/20250617210438/NEC-FC : SUCCESS in 6h 15m 17s","accounts_in_message":[],"_revision_number":47},{"id":"b6fe134b657bddb482d2526e278b2284e4ce897a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-20 09:37:10.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"4ca31cac5e08e550ef82dedac2e84624bbb401ab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-20 09:37:51.000000000","message":"Patch Set 48:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 916722,31","accounts_in_message":[],"_revision_number":48},{"id":"c95805b3373a80a26d9f2a570be8bae0f1afb60e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-20 09:39:54.000000000","message":"Patch Set 49: Patch Set 48 was rebased","accounts_in_message":[],"_revision_number":49},{"id":"89e02939ceebc0fedd6d958ab815478015c9d2f4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-20 09:44:43.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c56522f487b4a7f90c8e213555b7647 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a598bac31f954e5f9c154a8aef44eb73 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":49},{"id":"4f187a7483193380ccc8f470643e2412b7d56a96","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-20 10:05:52.000000000","message":"Patch Set 49:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/11a6cf4eb7cf4d96a3874ca822da9e81\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53ecddb65e9b48358af5cb93f445e7dc : NODE_FAILURE Node request 200-0000068507 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/56c0df0b58264aac80ca0465da43ca63 : NODE_FAILURE Node request 200-0000068508 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a0c034ff6e64c97a0dc8a9839fb1430 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3738956a20084437bfb3892a83ff54ac : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d3c80658f2fe4511a5fcf65cc352d032 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":49},{"id":"673de9eb957c2f4b8753601b3b3e05f0be4c0e55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 10:29:46.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed25db79e554b6891e0e7c995920e94\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38e71d0aac1447108fc3a147d3f8eceb : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea0ffca2b16445db8d62ca7ab7ae35dd : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2fd8d009a0294494a5533c969626a227 : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"f1779edb81db81cf452c08c9675b50948c77044b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 10:35:43.000000000","message":"Patch Set 48:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1009/.","accounts_in_message":[],"_revision_number":48},{"id":"b441af0a3c2810e6a6a27e666ae3b5dbeff89c8a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 10:40:41.000000000","message":"Patch Set 48:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1138/.","accounts_in_message":[],"_revision_number":48},{"id":"2feeeb059ab92e86f1d19e6460aebf64a9e7505f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-20 11:30:57.000000000","message":"Patch Set 48:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":48},{"id":"e70b98165dcdedd7b21fee9b2288863ff91b82fc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-20 12:10:35.000000000","message":"Patch Set 48:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-48/FC/ : UNSTABLE in 4m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-48/iSCSI/ : FAILURE in 2h 33m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":48},{"id":"2d338b3ed65475aa58631b6004125fd9a9c6bdcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 12:38:05.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ff84d34cef044fd9576bd4995c440af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479e3a7536e043c6b1b8c4bfa37e4cec : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3789567071141bf87e5f4c03ba10f29 : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80241f42da2349f1bef2c64746f3cb01 : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c308e4bbcd8406ba900e7b4262b7296 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505907a0a2c949728213313194643102 : SUCCESS in 15m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4132f4eae04b44d997de99ae8bb18930 : SUCCESS in 1h 03m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69476b2ed8ab4752a6ac580164a30b7c : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4eeda0d53ce842c58f52625be3460670 : SUCCESS in 1h 38m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3737014602e84df79cc46c9446ba5a88 : SUCCESS in 1h 00m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71bb1994d866420a9c193897a263853c : SUCCESS in 17m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37201785006b4eeea4db2ae8128beecc : TIMED_OUT in 40m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/102a033d70bb4988bf33ed695bd74c88 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c397e5427ea84bcf9c803e2c85fb4ec2 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04ac1169ceed4a7f8b674916f9e8093b : SUCCESS in 6m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b471eae7d36d49f2bfabefb163bc26a7 : SUCCESS in 10m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ac94cb53356413ca42a5d3fb6ff3f9a : FAILURE in 30m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9aa6d57668e340d680d0af0d5b769a6c : SUCCESS in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/345705b2a74649379995d6ba23bc0911 : SUCCESS in 1h 54m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aebeb9eb24fc4e5b88ac8f41e9f24a84 : FAILURE in 2h 44m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/663172c66fd44233a20fdea77ac99cf5 : SUCCESS in 1h 41m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0dbc802da78a4b4c84275a53fa5b3637 : FAILURE in 14m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/87f23ca8a43742d580b5bae07282bf06 : SUCCESS in 32m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc9d7ce968264ce49f8a0644faf8d718 : SUCCESS in 1h 07m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b0b80271f50f4b32903687e608270753 : SUCCESS in 1h 32m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b03aa3581c404c9fbef5a2b1da7bb0e5 : SUCCESS in 38m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e25bd64a099d4b808e2ecf5546e67d96 : SUCCESS in 1h 41m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/80f25ab5b0e643e88aaed8987a35f80d : FAILURE in 22m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fdb885611457457b926f3032123084f7 : FAILURE in 1h 40m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ea2f6f0043142aeb4488eb28a5deddb : SUCCESS in 1h 38m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a0ea27ee4a34a2b982d3bac25d0faab : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":49},{"id":"bf3a95a7b3d7532e8291200784367f0f59b831e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-20 12:50:28.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 35m 53s","accounts_in_message":[],"_revision_number":49},{"id":"57f153e9b24bd9648c0d481f6b76caf9253078a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 16:32:33.000000000","message":"Patch Set 49:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1144/.","accounts_in_message":[],"_revision_number":49},{"id":"cd0bc9e4ddf1eaa2cecd8d2d678fff99f54d9961","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-20 16:53:32.000000000","message":"Patch Set 49:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"b3de7f245edc42737d64440019b0df397720cc7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 17:19:30.000000000","message":"Patch Set 49:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1015/.","accounts_in_message":[],"_revision_number":49},{"id":"7a406d525c2f1bdc237a974dd6da4896a38365aa","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-23 00:50:14.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/49/20250623051424/NEC-V-ISCSI : SUCCESS in 4h 06m 23s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/49/20250623051424/NEC-V-FC : SUCCESS in 4h 35m 47s","accounts_in_message":[],"_revision_number":49},{"id":"b91e06369027e7c0ff0566aff338fd098495601b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-06-25 09:49:34.000000000","message":"Patch Set 50: Patch Set 49 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"af849348c94ce0e5c3ef5408ff1e21caba5e4519","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-25 09:54:25.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b5b064649b534eb0b9f933b354fa02f3 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bfda89cf95e94139b91ef404db962cc5 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":50},{"id":"5f43e052c435b35a9e8d22fba2f614911ad61895","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-25 10:24:40.000000000","message":"Patch Set 50:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ab66faee5d3f42d39912468cedcef54f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8ce3563630a47dcaa30a79b27af3852 : NODE_FAILURE Node request 200-0000069082 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/21e4dd213ea64749b45c9ee32b61c8f9 : NODE_FAILURE Node request 200-0000069083 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c1b86cdd31a489fbcb6f7c98710658b : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c573a54644ec46aa8a45212d6794bcee : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b093dca1b09748baaaeb04be5c456d46 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":50},{"id":"1d1692359dfac80336740121f4b71ebeeb66940f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 12:15:25.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1628fbbd5b04997aa65e68a33ee11e9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c64f683470da44fda5ad8da9a6ae0bb8 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/954f0aa6edbf40aeb8f65545b2f702d8 : FAILURE in 30m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19399fe6117b4ce8b62109f68b8b1adf : FAILURE in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"92e87869b4eb0b936355fc4383ce96d38c62d47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 12:51:00.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e93c1a379b604a4698badee9e74a3cda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde93a9d3b9449c089d54186bf4aaa2e : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07aeff9ff8814d248f4d420edfc6ad85 : SUCCESS in 12m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84043638ae3d45e3bd24efbd19e723cf : SUCCESS in 14m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19c7fdede94e4d719ac88cd80acc9843 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f7407b08864f8ebd113804073124e4 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e29d4358ad5647db81c30c4226fb7ccb : SUCCESS in 1h 16m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39277432b42344bd85bb965a74293312 : SUCCESS in 1h 02m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c386341352b84491a03d85590ba1b75c : SUCCESS in 1h 20m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8580bf5421c14585ab7194c8f258d5d8 : SUCCESS in 1h 02m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf956436bf244c2d8b357b64b8128169 : SUCCESS in 22m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c59b6338ad694a4c9c02424c7ce16a57 : SUCCESS in 22m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7593f5efc0174da691c5f7d8c1f0eed7 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfa04333d5774416a426bfcfc67c002f : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a9e62beb6234401ac74d45da55b18f0 : SUCCESS in 5m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f651cb637adf4db7baf1d1464ea69870 : SUCCESS in 9m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac43ae6063d54428b996dec3c180ff52 : FAILURE in 34m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c10470a451c7412f95f8312fd8aae2dd : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0167fbfe2f942e38280733fd5529326 : SUCCESS in 1h 39m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/009fb3baf6b94c9d808876f8e343ae4c : FAILURE in 2h 24m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/007a746b40b34ff68976d960b460c81a : FAILURE in 9m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d47f17d6511f4eef8112673ffd467db5 : FAILURE in 13m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a1ffa1d7a3374fbdb534aef7f1016caf : SUCCESS in 26m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5ae57fd7d164df694ad3def2a750187 : SUCCESS in 1h 12m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b577d7ca78b4e9389e4451574cea7b6 : SUCCESS in 1h 23m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/000a886afc444501823082ea21d6bef9 : SUCCESS in 36m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ea19fa41d2654cf3883bcddede42d2e0 : SUCCESS in 1h 29m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/feee771c46f142eda72d25f93ec2d98a : FAILURE in 12m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4da269436fe943a581353399366e5981 : FAILURE in 10m 50s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69ce1a2e200a46b3b49f1453cf924102 : FAILURE in 8m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88bd81e46aba41498209155e4730af0f : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":50},{"id":"bb395321666ec168243ef844f032f34581f060a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-25 13:33:35.000000000","message":"Patch Set 50:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":50},{"id":"b06ba8d49e4eb7e47843fdc174b777263b57c773","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 13:50:27.000000000","message":"Patch Set 50:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1190/.","accounts_in_message":[],"_revision_number":50},{"id":"dbb317a190ac8aff9564cb855f4ed86fafefae47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 14:40:40.000000000","message":"Patch Set 50:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1061/.","accounts_in_message":[],"_revision_number":50},{"id":"ef8dba98a88f04ea639c8bd3558f2a43584efa5f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-30 21:03:26.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/50/20250630234646/NEC-ISCSI : SUCCESS in 5h 49m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/50/20250630234646/NEC-FC : SUCCESS in 6h 16m 37s","accounts_in_message":[],"_revision_number":50},{"id":"dee86caef7580b96c3e78ae81739b7315dd38d54","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-01 02:15:21.000000000","message":"Patch Set 50:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/50/20250630215551/NEC-V-FC : FAILURE in 12h 30m 22s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/50/20250630215551/NEC-V-ISCSI : FAILURE in 13h 19m 27s","accounts_in_message":[],"_revision_number":50},{"id":"11732ae9d7e198bb8792dd46c52216bd816a062e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-01 05:28:15.000000000","message":"Patch Set 51: Patch Set 50 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"7df33024f39453e2a8fc73a098a2212d5bd4e3dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-01 05:32:40.000000000","message":"Patch Set 51:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":51},{"id":"3d04e8a04e9deb638cb28c46696196b19ab574f2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-01 07:02:32.000000000","message":"Patch Set 51:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b55d25be522848ab8d9fa63e12dd08c0\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16e97e0337a14358a316bd58f92c5fad : NODE_FAILURE Node request 200-0000069776 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2126321ae2424d6887601f05b905fa59 : NODE_FAILURE Node request 200-0000069777 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1829b179805d4d73ba91ac05b96e2620 : RETRY_LIMIT in 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/188d45295acb4e8299b183317fb6d952 : SUCCESS in 44m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5677390d9c9847869aede830b12e7854 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":51},{"id":"aa324ead7ae25030d0792b375e357dcb2695837c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 07:35:51.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24b510221deb4e6482adf5c031880e3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7072bab9a182406fb7da507d8e02b437 : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ab6858b19e444139e032237c32b27cf : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8344bbef6f54c5aa4c7ff93d34df791 : SUCCESS in 7m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2dfa12c33f44eb3a61aaead9a907573 : FAILURE in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ee5b94e1db46278b4cec5f52d9c1c2 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/95cad8670d1e4dafb647544df3fff34b : SUCCESS in 55m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a44c2d05b374e058549e79897fa0b6d : SUCCESS in 1h 37m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2dcb867ae1874644bdff61a2e37ba110 : SUCCESS in 1h 37m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a414720eed094254822289f6f63956a1 : SUCCESS in 1h 01m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e070d6a4a67b459ea0f6b1338f8a8cf3 : SUCCESS in 32m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d897b5916d546fd90fea7e152ac4039 : SUCCESS in 16m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9711235c24624dc48f520fff454c30d5 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b3385416e7944d1998f6ac0ca5b2185 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6eff5029bcd84667ab80e118ec3cf821 : SUCCESS in 5m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1262fc0a7f6f4c7ba7b1807d4563e21e : SUCCESS in 8m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/717db207bc404dd1b48c6442c9cf17af : FAILURE in 19m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a80bf33b21c4ddbaf4d0396064b59ee : SUCCESS in 3m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f83b957e548a49e7b9f7dc89061035fa : FAILURE in 4m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09a6efc272464ea5a46d3569dc91bc5b : FAILURE in 6m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c8f108e0b5a4239b12fa260f826f20c : SUCCESS in 1h 24m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33df8b6ecff445f492d9ffae33c77d57 : FAILURE in 12m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/68ce0b14bf884368a3ba89fe104a7d08 : SUCCESS in 28m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a037876da34941dbb548c34e9f0da8c6 : SUCCESS in 1h 16m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f95837520fe646be9665d12b8482d6c8 : SUCCESS in 1h 57m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3fc9c1490d214cb08a4cb4c388717f41 : SUCCESS in 39m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c8246ccdf6246b58505903a955ac729 : SUCCESS in 1h 36m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c5b221dbf7d24cddbcd557d3e34d75a2 : FAILURE in 21m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de6ce2b408f04d6880b898bfa1f20c20 : SUCCESS in 1h 58m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a77e72bf36954aa990c2745d18e2a041 : SUCCESS in 1h 48m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43f64b8c08b44914aa46e1685d047fb9 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":51},{"id":"9540222b690c7469274f34da030f18aee5df56ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-07-01 08:23:30.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":52},{"id":"eb9ebc14be48e13931d529bd2b796b3e693154bf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-01 08:27:17.000000000","message":"Patch Set 52:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":52},{"id":"46d3f44d39eab141fb56dbe0ea7488d8f2415302","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-01 08:50:27.000000000","message":"Patch Set 52:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/94a61c47ec7343a1aa59e48bb655bc77\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3884c3370a8d4f15a3837d0904b2e2a7 : NODE_FAILURE Node request 200-0000069955 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e266c160e1f6469690706f9fa344adcf : NODE_FAILURE Node request 200-0000069956 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46d80a38dac94e76bd39e355ea52c7ac : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd6a6a06a0af4663aff6263775b3075f : NODE_FAILURE Node request 200-0000069958 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86f3dfd68a1c4e4281c1a22225a8c632 : NODE_FAILURE Node request 200-0000069959 failed in 0s","accounts_in_message":[],"_revision_number":52},{"id":"0b5b67b3ff7b233521dd2c5ce18faf45a16b4d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 10:45:44.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2860aa245f04f4d8b2870e065fe80eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9967357d27544ae8ddab3558e8f99f7 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e210725eaae24028abef998ff7aaf956 : SUCCESS in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c768aeec34c7451c9f0ce19688aaffff : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58752ee37c7d43f3be8c9067d3da50f5 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf20a4b4b404a61ace7328f946c66a7 : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8187bd6bdf334c6a8fc4f920667c50d9 : SUCCESS in 56m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9bef93f95f04f3da25360fec0c44a12 : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a44ad061319f431eb57e01f67f10ca46 : SUCCESS in 1h 27m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc3e5ae6656a4ba1bd515be02030d592 : SUCCESS in 56m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0b9ee0d7aa4f23bdf72266f07cf428 : SUCCESS in 17m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/800c790251c6457b83f783802b292785 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a64da799165040619d33d1ebded1d4e2 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7091db017133438aac6f77de696970eb : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae051b1ef9f7420a87b6526c17b7c6cc : SUCCESS in 5m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e2ce0e08709430285b701c7ee30a69e : SUCCESS in 8m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f027787354841a18ae7d7ede72e8836 : FAILURE in 30m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b61502576cd427eb920f0edd0096db4 : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e3b618ac8124b1cbaad0d42ef612d0e : FAILURE in 4m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/957ba41622904dada85978c64ea32887 : FAILURE in 5m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dabe804ea2eb40059d0eb48675cd1213 : SUCCESS in 1h 31m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3b6d458946064841b6f04e21f83d254d : FAILURE in 11m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/89ed37317a0248e98d825a78262d672a : SUCCESS in 27m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/581965a132674e65b21daa6be4b3d927 : SUCCESS in 1h 12m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aa80848f2ece47278901ed6856c00223 : SUCCESS in 1h 27m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/870f9d7c79a24c4fba5fea124f0c959b : SUCCESS in 35m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7da07b9f3bd2413c8116962876be0324 : POST_FAILURE in 2h 14m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f11d18b54edc49a893acbad3d56c46d2 : FAILURE in 13m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e754a9f5b5274842b3d5cf0d10fe06d5 : SUCCESS in 1h 59m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1c1badfca3547a6925ddab6f72c53eb : SUCCESS in 1h 22m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3db00e692c9847fa8bcfdfda857f0eb5 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":52},{"id":"ccd66fa341b9862ed2af2eb586751eed16068268","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 11:22:41.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0902c043bb0a4912bad46e0268930e97\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb1a26e3119a4d1a9f11fa4d140f6b26 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d0519de901240cfbfa97ccb74678779 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dab445b881e435787ee34bebc1c3074 : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"a770b76fc56912166cad8506a66398c4404630e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-01 15:22:50.000000000","message":"Patch Set 52:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1250/.","accounts_in_message":[],"_revision_number":52},{"id":"04fe396bdc5086f473e042b0f4adde3ed4b75094","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-01 17:14:16.000000000","message":"Patch Set 52:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1121/.","accounts_in_message":[],"_revision_number":52},{"id":"ac5dc1fa6bfa398afdc96984b072dd21e558decb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-06 23:47:44.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/52/20250706234649/NEC-FC : SUCCESS in 6h 18m 02s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/52/20250706234649/NEC-ISCSI : SUCCESS in 9h 00m 51s","accounts_in_message":[],"_revision_number":52},{"id":"c7b6d7785f3398e7c83c10c3b702e6481d006b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-07-08 07:15:23.000000000","message":"Patch Set 53: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"acf7f0447afbf7a066a68267e05de8d97f501bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-07-08 07:15:30.000000000","message":"Patch Set 54: Patch Set 53 was rebased","accounts_in_message":[],"_revision_number":54},{"id":"1882b781ce5217269ec6fd52492a8a030d72d5a4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 07:16:31.000000000","message":"Patch Set 54:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":54},{"id":"6aaf9a5bb587e8b3213c169fc78bb5754cbce45b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 08:13:34.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef639d623af488984d87e96ea670133\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f597e2604424a88b7fe7c065ff4f099 : SUCCESS in 25m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/671509376b2d4f20a14f7549c8842a0e : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0faa63336eb443cbb8d73099945a1ade : SUCCESS in 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"543f8af28c12e1ec5330bd93f6a7c7884048d8c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-07-08 08:26:53.000000000","message":"Uploaded patch set 55: New patch set was added with same tree, parent tree, and commit message as Patch Set 54.","accounts_in_message":[],"_revision_number":55},{"id":"e388226654698516cbbcdc3a12dea110a8a75384","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 08:29:18.000000000","message":"Patch Set 55:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/253b3f283df84b1eb715456b8ab29180\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13d8173889c64f8c8275323b4933f165 : NODE_FAILURE Node request 200-0000071917 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9778a2e50dfd4fe5b39c2f460968a6b1 : NODE_FAILURE Node request 200-0000071918 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7dadde4838d94d1e9641c8afbe67ff01 : NODE_FAILURE Node request 200-0000071919 failed in 0s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f5b53b4f9274994a55ec828989377d0 : NODE_FAILURE Node request 200-0000071920 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba8e9076d9fb41f58f05a9017fc98748 : NODE_FAILURE Node request 200-0000071921 failed in 0s","accounts_in_message":[],"_revision_number":55},{"id":"c4032f5089b49b7b37ba09a484e23f0845e0abd0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 08:29:40.000000000","message":"Patch Set 55:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":55},{"id":"7db005704c6bcaedcecb6c8312bf6c73e6306ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-08 08:44:42.000000000","message":"Uploaded patch set 56: New patch set was added with same tree, parent tree, and commit message as Patch Set 55.","accounts_in_message":[],"_revision_number":56},{"id":"d2a0fbe0c7d0bbb8b4be9c523533a730f73fcd17","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 08:48:01.000000000","message":"Patch Set 56:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":56},{"id":"3d63306f8d2a8df4d38dc14d746de1aa7b92de3f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 09:16:49.000000000","message":"Patch Set 56:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ab4dfd1f3add46b68af256632916c2e1\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea68452175c44513a8dba20f1d36be14 : NODE_FAILURE Node request 200-0000071956 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5ed5bec3ba0461bac912cd52041bbec : NODE_FAILURE Node request 200-0000071957 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/231b22802c994776bbf26fe68e9a0961 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6cf8ffe3af14dcd9bf3dbf165e905c4 : NODE_FAILURE Node request 200-0000072148 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86ed143fa40549719dbf6ae0c75b0e7d : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":56},{"id":"2315aac17e79ba7245e52fdde076a14a4e721a0b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 09:40:56.000000000","message":"Patch Set 53:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1312/.","accounts_in_message":[],"_revision_number":53},{"id":"e7c336486d487f63376b2d5266591dfe11f132be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 09:50:14.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5c2da8e62540029cb3a5b9f19596c6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bce01c35e8b44b8959e0ded3a5ac0ac : FAILURE in 33m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53982c210a2e4a5d965eb7bc4b071f3b : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b25535c1532942ef86fe4bc0db70501c : SUCCESS in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"1500bb8d64129a0c3a8c69bfd3e7be91bf34de0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 11:28:03.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a9a2f439fd46ab928c2b0969aad9bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6299b388fa4d4685905985a25217e173 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d622cd489244f2ba093abe7a2fce952 : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eac5ea402ce74ebe896df1e6f17087c4 : SUCCESS in 10m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4c459e2e561420888b1cfc7ed80ffa7 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28708cdb659148a0a3345e8f5740c5aa : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c27e4754884d4bbbb10328d6694e8fd1 : SUCCESS in 52m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d517468c76e441383dc20854c53234a : SUCCESS in 1h 04m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efe074f51d844c08b2db5a906f4b7abf : SUCCESS in 1h 32m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7f8abce0b46441bb331939113d61e2c : SUCCESS in 52m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99e1e54153654a99be2405c107fdc64e : SUCCESS in 18m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3571fd7cd7e346ad9b0f75ae508a0c0c : SUCCESS in 16m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1169bc56e1946339d948705188106bf : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfcbf705010c4dc9af2976db9d7d677d : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86089618db704f41ae966985bab6f783 : SUCCESS in 3m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5ec3bccd6ca94f619fcb8d4e78e874ec : SUCCESS in 8m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb919a8760ab48bc894f93b482c89a1c : FAILURE in 29m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c959bf61685740f58c58558e77fd6fe5 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af62e9c9241b4e6aa9aafb426c7348c8 : SUCCESS in 1h 47m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ddffc1a32c544a4784729fd5f2c7a870 : FAILURE in 2h 39m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0ac951baa9c423c9757d65c8bdbb5c3 : SUCCESS in 1h 54m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e0b49de7515e431d972472312cd708a0 : FAILURE in 15m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/be5972a10a3b49deb0fc2057ba3c2b13 : SUCCESS in 18m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9469a9184a06440a91a3f9bf812ce20a : SUCCESS in 1h 10m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5a7393eeaa547b4b709ce9a9231a9c4 : SUCCESS in 1h 23m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9799525dc24248f2a142ca28bed26a1c : SUCCESS in 33m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/725f7c31479140419347f888416065d1 : SUCCESS in 1h 14m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b90d573c0b49446bbe8a6a71b111cfdf : FAILURE in 22m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/752f7099c2564ee1b09b06615c4d4083 : SUCCESS in 1h 24m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c86e50dad14b45b68986e68cd8e0172c : SUCCESS in 1h 36m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16d8de6e9f5a49dcbb7e361321159a3d : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":56},{"id":"3c2b35997490ec9286f9f0c50f7144b069446bf1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-08 11:36:22.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":56},{"id":"d62b88bcc54cda39c56d4e8d78dc17df3b2a7c0f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-08 13:12:48.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 04s","accounts_in_message":[],"_revision_number":56},{"id":"98783aaea18c77ee96ebd5f14b95938762edfd25","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-08 14:12:06.000000000","message":"Patch Set 53:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"9bf23943b25fd3b2c84a94962bddf14328916ae2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-08 14:20:17.000000000","message":"Patch Set 54:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":54},{"id":"43b4e0f358ac2cb2c9b47e3891dec2f4fd2d242e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-08 14:21:41.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":56},{"id":"aed0ad099a6dd5bea7747a47a98f6838e269776d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 14:40:38.000000000","message":"Patch Set 56:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1188/.","accounts_in_message":[],"_revision_number":56},{"id":"5a025b0a7bb15503deca4338f7e0c9105bfde98a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 15:31:26.000000000","message":"Patch Set 56:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1318/.","accounts_in_message":[],"_revision_number":56},{"id":"f85255df765c702ac3438b3b0f1dfa2ba0e2a5a4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-08 17:55:49.000000000","message":"Patch Set 55:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":55},{"id":"e9893d0e4e665d3312b56bd9749d0aa5d83f8d2b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-08 23:27:50.000000000","message":"Patch Set 56:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"94f05c1ad42d7b6135dad83d9eee3492f099d062","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-14 19:07:48.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/56/20250714182827/NEC-ISCSI : SUCCESS in 5h 49m 11s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/56/20250714182827/NEC-FC : SUCCESS in 9h 39m 18s","accounts_in_message":[],"_revision_number":56},{"id":"09672e772c966401fe4c6134e97dfeb2b1ddb3a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-17 10:09:32.000000000","message":"Patch Set 57: Patch Set 56 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"23c0c2ae2e308235361a7476f2ada27531e36a2c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-17 11:28:40.000000000","message":"Patch Set 57:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3d280259110547bf972226f656e5a71c\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d11ac203de564595a1c4815546d3c8d6 : NODE_FAILURE Node request 200-0000074077 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8cf97b3f5f14e43b602d426fa296705 : NODE_FAILURE Node request 200-0000074078 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4bb899d7a5ec422295fc3a0dd1e83d7e : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9523caaddefc45a98fda8d84c6cdf2ca : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8314e0f0ae8e4077b1845576e7a413f6 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":57},{"id":"59d80381b9ae0fbb7c219702da3c3aa162c1ad36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:14:00.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9788932bb03b41e6934fdfc9d36c4fac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28e351e42f9747f8ba5da6cbb323734b : SUCCESS in 23m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/688c81da70a345f9a24c32a963ad5155 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/adf5d45aa44a4dbab3281e744a60092e : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"ce32e353ae76149abbe35a9b31ead5ee663b6da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:55:17.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0982cbb3b846d59fcb0c223249a3e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f179ac37c4f4e7ebfebf69918e8b51b : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13aaba11a42949f8ab99bcc10dee38df : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80cc6dc212034de7a2853434f49e7373 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fedb3af51e04d86af4baaed6cbb8ee4 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224bbe160e7640ffa2113007767a3388 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/def3f820dddc4c7e8686dcb623765270 : SUCCESS in 54m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49f8311fb73941d784b93a40cff10329 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/739c97fbc8cd4473b47bd1049f0fcc82 : SUCCESS in 1h 20m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27be21db40024f78a9a7e5e796550a12 : SUCCESS in 1h 07m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e17830ec0c49fab106380b9457f12d : SUCCESS in 13m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a202e62f6eed4f8884013b85a5f354cb : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/981f792b82a64909964d7f2e0a809ee3 : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54293f1667884e5e863172b704a6d41d : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7ccad974cef476986bdf49e768bdf51 : SUCCESS in 6m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8bd4c18a559a41f8acd43d3749f3f05a : SUCCESS in 9m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9baa560d5dfa405b8528126109148c83 : FAILURE in 36m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89f8669312b848799e476df8a5716cf7 : SUCCESS in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49d5ce5d303847c6b19ebe01a9d431fe : SUCCESS in 1h 54m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5396b43f5c51437383f5c12781a86311 : FAILURE in 2h 28m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5d939bcbf8b43678d4a9b3b8e19d398 : SUCCESS in 2h 15m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e87784b2b8574cf2aa633862bd86e62e : FAILURE in 17m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/882d3022e1a84fdcb74e38a84fd9db4a : SUCCESS in 27m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f467ea65be934310a38057f8a78f4044 : SUCCESS in 1h 08m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/740e7edb2cbc4761bd52575edc08eaee : SUCCESS in 1h 49m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/15a53dbedc5c416a952f83a40a3b3520 : SUCCESS in 36m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3e72cd5a51cd4c909a5ce69e7ea9fc99 : SUCCESS in 1h 37m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/08bdd35da02b4458abcbe01e0daf5403 : FAILURE in 21m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73d35f9fae1e4a14956eee4117bee502 : SUCCESS in 1h 41m 14s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ae3fd1eca84c14b18def5b99a34371 : SUCCESS in 1h 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af37baa3b4a44d1bbe46f6f9aa878436 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":57},{"id":"6cd777226d9090594cac5cf9a3cee29be90f47ed","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-17 15:46:36.000000000","message":"Patch Set 57:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-57/FC/ : SUCCESS in 5h 32m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-57/iSCSI/ : SUCCESS in 5h 36m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":57},{"id":"414c26039fdfd0541ac69a326c6afbb043c3732f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-17 16:32:48.000000000","message":"Patch Set 57:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d710b8f5feac4aaeb2751976d5b794d5 : SUCCESS in 49m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e57f2eb523d04ad88c1da12b54fdafad : FAILURE in 30m 04s","accounts_in_message":[],"_revision_number":57},{"id":"3b5293c0b9a104289a569b1034df7653f825ba37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-17 17:00:21.000000000","message":"Patch Set 57:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1246/.","accounts_in_message":[],"_revision_number":57},{"id":"9ca91e29cfa09eeac3555180249d0564a4475baf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-17 17:01:19.000000000","message":"Patch Set 57:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1378/.","accounts_in_message":[],"_revision_number":57},{"id":"4fd79f22d90a006106a29bcd8b93c47418fa2cc7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-17 17:55:41.000000000","message":"Patch Set 57:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":57},{"id":"30e5f9c410ea0d5e5f29704969f6db92e25b14d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-07-18 07:25:16.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"090f6168ed40bb19c9d1656a21970e53142b59c8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 08:02:06.000000000","message":"Patch Set 58:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b9c1c7cd1baa4543b6df8fe3ec2d9c44\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53c4e5ae1c9e4fa8bd0b74d0a79d0326 : NODE_FAILURE Node request 200-0000074669 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64cb495962c5406eb72334e4f17be7c4 : NODE_FAILURE Node request 200-0000074670 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b7b6c7d493d48039b486c935b1bea56 : RETRY_LIMIT in 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ffc4a4e8e30247369711700f17954de7 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e12001af1c5b4607b7b13f67114bbfdc : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":58},{"id":"5b1e19c4be5cde02ceb987d0a0e889c1b249350a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 08:16:40.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd8e5bf4eff46f18d9bb8506e6a228e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/add65598d140412d85b7de6bb9c300f2 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39e333167c704a59873427c6cfed0ace : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54ef31669b454b4ea40e14e5b9333041 : SUCCESS in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"18a4374b86ec5761bdcc7ebea87aa572affa2cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-18 08:33:19.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.","accounts_in_message":[],"_revision_number":59},{"id":"c1b6eb82f23c63da026f233ab3eb8404134055f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-18 09:20:21.000000000","message":"Patch Set 58:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-58/FC/ : UNSTABLE in 14m 9s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-58/iSCSI/ : FAILURE in 1h 54m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":58},{"id":"7dc1b6d3a7dd64252ed13d3261a46372bc187075","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 09:23:25.000000000","message":"Patch Set 59:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e2c0f6cdca274d9cbfa6f0e86509b933\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da21253b2c3e4ca8a5da526e62ebb2cf : NODE_FAILURE Node request 200-0000074740 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a7d91277cc7499c82611cf6d5187484 : NODE_FAILURE Node request 200-0000074741 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f9f5e9988cfa424f8f8610befd8afcbe : RETRY_LIMIT in 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/49b5d66dfd6e45f6af96ad60bb2c0b2a : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/db2843c52fa84392baf9ffa4d45b8053 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":59},{"id":"9c73a66345df9448c5d1ca1621906898328a9520","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 09:31:31.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47df3e6e2b794f3898bb6f464553e479\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5508869c4fc438fafd18d71669b7b21 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/385f14a546b5476582a1aa62d6d8d7d5 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/957b9f2de09e42a1ac1356694db56c82 : SUCCESS in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"ee6d9a5ab1cb884652b4e58291553ea1ffeafaf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 11:15:10.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7daa38427b54e0baffb2104b5145207\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ee36ade22b458291fb1d2b84bcabc5 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dfc053e9a334a11bdc388483484cf0d : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/010cb890ce4a4daf91d4093da036abbd : SUCCESS in 10m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b3b747ec908499581099a8050cf0359 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b272e2923440399c46793cec2850b7 : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/9aaea75573474e2db8c18c1e40e448be : SUCCESS in 1h 02m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78350b06f49148ec81023c9967e94882 : SUCCESS in 1h 08m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1d20bd88b534cd79443ef399ec37550 : SUCCESS in 1h 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c11389306c2477fa1b2e0b925e79501 : SUCCESS in 1h 04m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cadbf7ffed3442f4844f1faf38b97ed6 : SUCCESS in 15m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5b437fccec34305bd22073088a98cd9 : SUCCESS in 23m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/efb0dacb6e3f486683feb062f4db5c9e : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/99c24d5b53d2415dbbf6207370980d57 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2a8da1556a44212a81cd1b96897dfdf : SUCCESS in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d49dc15af084bf98e352df25e545411 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59b5027ae65d48759c862b9790b560c6 : FAILURE in 41m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80bbf0988f9446168161e75973836c5b : SUCCESS in 7m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a5bd9ff2c1a142dfb7d459f1f442d907 : SUCCESS in 2h 07m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1ca8c71ec9343bb9b59828a9e1f2c14 : SUCCESS in 2h 35m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3535bb44c39e478a97453c8eae42508b : SUCCESS in 1h 43m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4bab810229e246a6bafd17c456737c03 : FAILURE in 19m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/86aed7f5ecff417399c165fa74425931 : SUCCESS in 37m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c2babaa5732a4e41a7122c9bab09f976 : SUCCESS in 1h 16m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c5afef72b4d49cbb5a1c3c4bcd92f8a : SUCCESS in 1h 47m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c14cbf5e77f44c079d2ecd6b2a2f8a6c : SUCCESS in 49m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c7ad25b8bc64e9fa5f1cdd9ac922cea : SUCCESS in 1h 30m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0965ebfb41304e1282b06a7b65a309c0 : FAILURE in 18m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57c4164f440f476ea2c8e6dcf6414e86 : SUCCESS in 2h 06m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0f072a2df104157940d9b5f7111fef8 : SUCCESS in 1h 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f1baeb3aeca49a08d640a10d89aab52 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":59},{"id":"3714834b555b8b0ea04fe988959e332dbe206d45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-18 12:14:38.000000000","message":"Patch Set 59:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1265/.","accounts_in_message":[],"_revision_number":59},{"id":"d8b80a42d0e21f98202974a92690beca2514b4d9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-18 12:17:06.000000000","message":"Patch Set 59:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-59/FC/ : FAILURE in 3h 43m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-59/iSCSI/ : FAILURE in 3h 28m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":59},{"id":"3306a122f7db6f4d73e23c3184d0dde939f89cec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-18 12:29:49.000000000","message":"Patch Set 59:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1397/.","accounts_in_message":[],"_revision_number":59},{"id":"78b2a88cd0efa08130adb6f8fff6b025ab34ef62","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-18 12:45:29.000000000","message":"Patch Set 59:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8a6b72ede81944d7816704178bc1b8e6 : SUCCESS in 52m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5087816deb4a4bfbbf056b31b49e48c9 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":59},{"id":"2f167bff885cda9a557c1de3de249c02c044c824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-18 13:05:41.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":60},{"id":"3fc583621463d3b605e38129727ed77b151b7126","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 13:44:54.000000000","message":"Patch Set 60:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a2e84f4bd6d240a08c32027551077e25\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee706dfaa56c42609f3b095c1d61ba29 : NODE_FAILURE Node request 200-0000075122 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7e481217e464715b6445ea44d1e3cc6 : NODE_FAILURE Node request 200-0000075123 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20087c06e07d4a6aa43616b40b884b1a : RETRY_LIMIT in 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08937ab8dc7e4effa5da5ba1f22512e4 : RETRY_LIMIT in 35s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c5dab12f93b4fd0b5d0cd7919bc3999 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":60},{"id":"adc4abb09c8edeb1f988398b7caa303a58b7b9b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-18 13:50:15.000000000","message":"Patch Set 58:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":58},{"id":"5618b1e18be6c06ab224fe9f15016f6625f18a0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-18 14:59:15.000000000","message":"Uploaded patch set 61: New patch set was added with same tree, parent tree, and commit message as Patch Set 60.","accounts_in_message":[],"_revision_number":61},{"id":"cb9dbc2c4ebc7000d395e78725dc095b39eb463f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 15:36:14.000000000","message":"Patch Set 61:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b725b9abec3d470ebac7823cd5cc3e18\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4bb8bff91a8c418c9131a024b5510e78 : NODE_FAILURE Node request 200-0000075381 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1e18aed589844aa0bf05cbc401a199a6 : NODE_FAILURE Node request 200-0000075382 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/076e99b6fa59410b8360c80b62cf63bb : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1a470d129cc411297b97d24f8a128ac : NODE_FAILURE Node request 200-0000075384 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b96061996e9f4653a82bf9833d012a8f : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":61},{"id":"5ce78a23356beba410d168c19a7a75baff86bb69","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-18 17:02:25.000000000","message":"Patch Set 60:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-60/FC/ : UNSTABLE in 1h 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-60/iSCSI/ : FAILURE in 2h 59m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":60},{"id":"66e08d497cc045a16b3dadf6c150d326b0b1f832","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-18 17:36:40.000000000","message":"Patch Set 59:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":59},{"id":"4abf0759be92c0ed49bbfac6cb32574411135ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:52:24.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e09a84539bf45d19ca0d9893f777132\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af592f97b4749458e2b4bd5f03354c9 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b94053b01bb14220982c368350f91083 : SUCCESS in 12m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac248f9c83bd406e99c3a9c2dbabaab8 : SUCCESS in 17m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dba2cdb8ed97468c89516f2d68a8bf75 : SUCCESS in 23m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb735a7adf1442e19814b979beaf028d : SUCCESS in 17m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/32a37771d18b440bbbcb246515a604e3 : SUCCESS in 54m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08600e5cdd3e4b61b9b11edf3adaa665 : SUCCESS in 1h 07m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33391ba4b04144cca8a9397d31b1858e : SUCCESS in 1h 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50a5d30416ee48bba88a443f0045bef6 : SUCCESS in 1h 02m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f89317d36dcb419eb27897cd74a2432f : SUCCESS in 16m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d5ef3839d354d90b446244aabb63ef8 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/81eb7d9295414cdf84cd029f892d946d : SUCCESS in 6m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc2526140cdf4f50a51f3d8e95a0ef47 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89b1006bb5ac4110b445bab54b60e51f : SUCCESS in 5m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/562c5c9460fd405289fb5b78e71e39ce : SUCCESS in 9m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4afc090d22d04abc847485191e722ebd : FAILURE in 36m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f19144d421db4ed799f7be7e460c0ab0 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a1306331262493294738834ed48aea3 : SUCCESS in 1h 54m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/79ef6031b6144338bab49a74cae95a81 : FAILURE in 2h 31m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/58b74f3fe49149ea955ce98229b03a4c : SUCCESS in 1h 45m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/35d4e927dc7146e7a14b1464b1e50964 : FAILURE in 15m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f04ca3a7609f40dca97389f8f83bf720 : SUCCESS in 36m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/857fdd5cd266413783ba109bd4299bfc : SUCCESS in 1h 19m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fbb34d3ae8a54e6097dd9d7f00ae85f8 : SUCCESS in 2h 43m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a51e3148a0b4e26bf9f04e53c90511c : SUCCESS in 42m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d4b1ec736ea44babbd29a073128d1618 : POST_FAILURE in 2h 28m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bfcf8406e58e417ab8f78ae3de511f36 : FAILURE in 16m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fd4d3f7e7454d3ea14ca30d0b49b92d : SUCCESS in 1h 24m 46s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23c10e653ea14da187400ba691f0bae4 : SUCCESS in 1h 33m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b2a67d16ca3495e988a29ec12c7291a : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":61},{"id":"f1d8d0842e2135091ae52f569757542a8cc7bb56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 19:21:22.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f4fc9542c54523b461ac71ddfa898e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1afd3a92bffb43dca9eb2afe3f6574bf : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e46b02079f5b4648bbc27e89fb57a184 : FAILURE in 26m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c870e17b68a48a988e39217387cb403 : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"53b205d663f19031dded6752dac05f7a868608fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-18 20:42:01.000000000","message":"Patch Set 61:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1274/.","accounts_in_message":[],"_revision_number":61},{"id":"17d703671b684f8fce5cb563c21e5934fa0cd8b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-18 20:42:45.000000000","message":"Patch Set 61:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1407/.","accounts_in_message":[],"_revision_number":61},{"id":"5d04408a5c00901ab07cd801e3c8f7149e597c8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-18 20:47:10.000000000","message":"Patch Set 61:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c66f3900a9944706b486d6b561494ebb : SUCCESS in 46m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/98f0337107334f3b85d7801c8af28483 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":61},{"id":"1b5f62c301ab60ae0d5396995548211cbc6c7e1d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-18 20:47:29.000000000","message":"Patch Set 61:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":61},{"id":"3eb4902061fc6d709cc773171fa65bad2426b68e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-18 22:54:42.000000000","message":"Patch Set 61:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 48s","accounts_in_message":[],"_revision_number":61},{"id":"1f0c0aad6c5a6a6db2ca1093ccccaf00154798aa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-18 23:58:03.000000000","message":"Patch Set 61:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-61/FC/ : FAILURE in 5h 43m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-61/iSCSI/ : FAILURE in 5h 8m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":61},{"id":"cb89b0a6d1e59e7406cb93a27058ea52ed58ee32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-19 04:41:01.000000000","message":"Patch Set 60:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":60},{"id":"e4a6325f7f4edd039f5ade1f02779072657de27d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-19 12:45:11.000000000","message":"Patch Set 61:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":61},{"id":"f95f967427efd07a27318575f35e5c505364d33c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-21 12:03:14.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":62},{"id":"1ae2e6493a6ac847f453903d1138859af6a2289c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 12:54:54.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd721a9f83a4916bb3f18e1d23c0e29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ef806967c12486b8230edc2412a7a9d : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66613c24f6434a079d8b3be8b83be076 : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a822739fa954af59566d84be86af9c5 : SUCCESS in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"cb0434b2732f218258a92da0cf7b20c261df5987","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-21 12:55:28.000000000","message":"Patch Set 62:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a2e4f95ea31146dd85e64e4252d69fa9\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/baae02caf3b6478b8dd9b0ac906d3464 : NODE_FAILURE Node request 200-0000075660 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3bc89746f5c241ab8456d1a7331815a2 : NODE_FAILURE Node request 200-0000075661 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df4acafb58904e2ca5ae64a051bac110 : RETRY_LIMIT in 31s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8bfaefc68a6b41edad603943328b2ba1 : RETRY_LIMIT in 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c3b479fee194477bb454d93a7e006b3 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":62},{"id":"4b58f46bb430451fed748ded2cd5ddcde8dcc3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 14:33:19.000000000","message":"Patch Set 62: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b379786d103d45c99f3d13d29626fd28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7788cb2988fa483883024b44b0f87c88 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d895a10b1bb45f0b37bdefa09696789 : SUCCESS in 12m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a96fd419af94df38b39b3944663bf50 : SUCCESS in 10m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f671370c6d44fcfa46289a2d55c07a7 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b8159643c744f7a828e94bf66d6f6d : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a4d2ac9a467d4cd0a1fd60a7150b5b5c : SUCCESS in 51m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/191398aa375d40f4b9f95abc0a421afd : SUCCESS in 36m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fad1bbefb62b4cc1a4c90f15ba5ea8c2 : SUCCESS in 1h 50m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47072400210b411aa0863fa98e0cff80 : SUCCESS in 56m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7f62e86e65451bbb90d856b20f9dca : SUCCESS in 12m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d9963c50e10475f8df3feea5495ef32 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b002328814d7477e880c75b71325a07d : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17eec7898fc043869023e273a157421b : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/862fd0e7a95b4eee8e9c59daaf6f9bb7 : SUCCESS in 6m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e9976c61f604b9786551fc2383694f6 : SUCCESS in 10m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51f35816e9b54d29b33dce94c9bfddb5 : FAILURE in 42m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/989e3815da5749e7b15e9416fc303712 : SUCCESS in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55a8c794fd8942699bd83ae5dee17ec1 : SUCCESS in 1h 42m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ebb2f776c004717ba9f75a8ba72c3c2 : FAILURE in 2h 24m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/674a4c13d31147a286ae7de02adbe61d : SUCCESS in 2h 02m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/956de15a8b9b48508115c2f201d2ffd6 : FAILURE in 21m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c5df0eb629454e218d6a9f95c45ac313 : SUCCESS in 36m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e34f08c6b22b4262aae35378a3fd1c28 : SUCCESS in 1h 24m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fab08a5bb3ad4d1480d3b3d498e4e2f3 : SUCCESS in 1h 40m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d7dcee3928024c30b5271d02301201dc : SUCCESS in 40m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1e2a7ec3bad4cdc85c6d8e5b4f099b3 : SUCCESS in 1h 26m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4542df799aa64ca0af143727c23fb5c7 : FAILURE in 22m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f9749d88e114acfac6d881a0e735350 : SUCCESS in 1h 28m 00s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/848f68b9b8084241ac55224d8ddf8268 : SUCCESS in 1h 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83da2a5631174849843ed908e6f8bcca : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":62},{"id":"367be7ca1660dc54a26471d2f77f56d1608289a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-21 14:41:49.000000000","message":"Patch Set 62:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_07_2025_14_36_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":62},{"id":"7ba2a7a233c9bd5769a8a0f683cebd7221df47fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-21 15:01:55.000000000","message":"Patch Set 62:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 29s","accounts_in_message":[],"_revision_number":62},{"id":"0fe80167e006bf1f98fb9e96ea494f4b685ecc12","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-21 15:40:58.000000000","message":"Patch Set 62:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-62/FC/ : FAILURE in 2h 40m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-62/iSCSI/ : FAILURE in 3h 37m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":62},{"id":"5b1cd16eff248c700d2fdb9dc91c08b812e6471d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-21 15:46:28.000000000","message":"Patch Set 62:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1287/.","accounts_in_message":[],"_revision_number":62},{"id":"b2601afcfef39ec5e63ce68adaeb11751f3920fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-21 15:47:47.000000000","message":"Patch Set 62:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 34m 37s","accounts_in_message":[],"_revision_number":62},{"id":"7e354b46c8c96e04793f432d60a0defa0c293aee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-21 15:48:49.000000000","message":"Patch Set 62:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e7175a967fe1421fa0c22ae7441866be : FAILURE in 48m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/904d6b9fe967452bb2eeed9a6e29f8d7 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":62},{"id":"359044a475a0ed79d1ad3d75866fe05621b78493","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-21 15:52:30.000000000","message":"Patch Set 62:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":62},{"id":"21bbd5c32bda58f4dcf093609c924351b9568841","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-21 15:53:30.000000000","message":"Patch Set 62:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1420/.","accounts_in_message":[],"_revision_number":62},{"id":"693bf155454e7a1a5f083b55702aca653e8170c4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:16:28.000000000","message":"Patch Set 62:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":62},{"id":"d3d5bbc007a091824bb68e6e9ddb4edc682dd932","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-26 15:40:11.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/62/20250726191655/NEC-V-ISCSI : SUCCESS in 5h 05m 42s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/62/20250726191655/NEC-V-FC : SUCCESS in 5h 23m 13s","accounts_in_message":[],"_revision_number":62},{"id":"ea68f2a6a2f9f0f9519de6d9a94c81c58f4439c5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-27 02:38:10.000000000","message":"Patch Set 62:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/62/20250727052501/NEC-ISCSI : SUCCESS in 5h 52m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/62/20250727052501/NEC-FC : SUCCESS in 6h 13m 06s","accounts_in_message":[],"_revision_number":62},{"id":"53aeee69a46325a4a25edbe1fe0aa3cbef528f79","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-07-28 09:44:07.000000000","message":"Patch Set 62:\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"991f72f6a02c4a420d2f8a3306b3964d281a8f5d","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-07-28 13:40:04.000000000","message":"Patch Set 62:\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"bfca5a544d1f09ab2c2b77899c6c23e526938487","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 13:40:14.000000000","message":"Patch Set 62:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/49/","accounts_in_message":[],"_revision_number":62},{"id":"bcd4a3293d42dc52d339d1ec9fbb30f7305a86a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 13:49:47.000000000","message":"Patch Set 62:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/62/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/62/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":62},{"id":"faac7557feec379c21fd0d9d93eda0d4d3d1e25c","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-05 06:38:18.000000000","message":"Patch Set 62: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"b07abadd86df78f578764c215476dbb78e144e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-05 07:11:59.000000000","message":"Patch Set 63: Patch Set 62 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":63},{"id":"33e73e35a56b8cb8cc59ef054d534e50eaced816","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-05 07:43:41.000000000","message":"Patch Set 63:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/337f0c1975a0407a88b4e283292cd648\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd4ee9021d024a92bfda3dd6dfe099d3 : NODE_FAILURE Node request 200-0000077304 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f572762c54b24244aaeee27976f2443f : NODE_FAILURE Node request 200-0000077305 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17cfa599d9bb41beb67bd78b1e475064 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61c18927f41c425f997435273c79be5a : NODE_FAILURE Node request 200-0000077307 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/586a1d9122e94d339caa8e9dde72cf44 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":63},{"id":"b003d5f015633500151a20c34db38db6a4a139ae","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-05 09:15:57.000000000","message":"Patch Set 63:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-63/FC/ : UNSTABLE in 10m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-63/iSCSI/ : FAILURE in 2h 3m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":63},{"id":"c7af15f49e45f77cf2d1c0005fe7837620b69d20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 09:40:10.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d1b13a78ba45b98aeac8249af2be95\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3009f0386464db492d44adb9450a0e2 : SUCCESS in 26m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49882d2b57c24bee889c10c3ed39431e : FAILURE in 32m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8913a0a840954ccb94f76dc440fb24e2 : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"11ab916b3b8acc0596751dff4c3b8749cb807358","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-05 09:40:40.000000000","message":"Patch Set 63:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/91/","accounts_in_message":[],"_revision_number":63},{"id":"78c0922ce546c6739dd318ff2ae040fb57ee3edc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-05 09:40:42.000000000","message":"Patch Set 63:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/63/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/63/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":63},{"id":"f98964c6d9b1ba33829baf43aa8dc3e106dd3628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 10:02:30.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8167ace0c1634fa880a1d4150f4f1277\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f77d9817f14af0910a0e78d413adee : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2a9d7e02aaa40cb9e7dc83f4d789bd3 : SUCCESS in 7m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1632ec32586642c69beca8fcfccdec7b : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9274f3ad2eb64bd3893ef22f53672746 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112105cd99c040a49bc8df316d8e391e : SUCCESS in 14m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/648b9a855a7848efbcf9597cacbf9a07 : SUCCESS in 1h 01m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/846b558ab9164c76b71a2f3299a61e03 : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8a47ebb35644e989b9067727ca2fb4b : SUCCESS in 1h 23m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bafefc9c2bb479093b5895ef927f1c9 : SUCCESS in 1h 00m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a5aa044afca482d8e18407b8913cec2 : SUCCESS in 16m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7887fbae313f4558813c0baecf808349 : SUCCESS in 17m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2fb2678203b54c8a9f32efd747e1e72b : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/742edcec56444839a35585df4c939ebd : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13d1e7554b2e4252b6cce838d8098010 : SUCCESS in 5m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa0893f2b9d543b8b2270319ab0865af : SUCCESS in 8m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2f390eff5ea241eabe692232acb87fb7 : FAILURE in 30m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6585454a018941978e1098cfb1895f2e : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e28532d0909040cabf7eb8e6e6b1b2c6 : SUCCESS in 1h 54m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/236d0e256d294c23a8b7c165cfbaf9e6 : FAILURE in 2h 36m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a1c29fa2e3e415c8d7cc538b1392b32 : SUCCESS in 1h 52m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9d1ccaeb3e3140b2b244afa25df24eec : FAILURE in 19m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/90d05fac3ce44fddbd34c2678570e503 : SUCCESS in 35m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/65be8fe145e84239b84bc60c4b33f9cc : SUCCESS in 1h 10m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c80235cdc204b2481bda355f8c509c4 : SUCCESS in 1h 34m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3ad85604d3954bde943da62bddc95762 : SUCCESS in 39m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/769eb9395a0d44daa81cebe508ba8558 : SUCCESS in 1h 30m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a978043e946c42ef9b1a0ebea2e90db8 : FAILURE in 17m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f9aa9f4264f4dbdb750c9cdc1e27b90 : SUCCESS in 1h 44m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f82bb037acb946c9ac3594865ab6baba : SUCCESS in 1h 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a271f57c3f42cbb5238b7026533b3f : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":63},{"id":"bed8b0e93cb4b6cb520f2f8fd4a0938f67f4516a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-05 10:12:52.000000000","message":"Patch Set 63:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_08_2025_10_07_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":63},{"id":"882a3ae16dd87ceaac540a3cc80802e246846b66","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-05 10:14:36.000000000","message":"Patch Set 63:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/938771/63/upstream-check/cinder-tempest-plugin-ontap-iscsi/38a3f70/ : RETRY_LIMIT in 4m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/938771/63/upstream-check/cinder-tempest-plugin-ontap-nfs/a9b159d/ : RETRY_LIMIT in 4m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":63},{"id":"8e3cf79540bc5a79d13623d685d431159c6c0048","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-05 10:37:19.000000000","message":"Patch Set 63:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 37m 07s","accounts_in_message":[],"_revision_number":63},{"id":"f756da40ed690e598c067a2570cd7294af3cf45b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-05 11:09:58.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/39e6302161f94549b329b4a5e19e0146 : SUCCESS in 48m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6afcca170bb04000b2b53e67d3da0443 : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":63},{"id":"f3bfa804a77f768b31f67852a57d6e6d8da112bb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-05 11:13:50.000000000","message":"Patch Set 63:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1359/.","accounts_in_message":[],"_revision_number":63},{"id":"37fdfb8b443f233ab5b98636f04b70106534ef97","author":{"_account_id":24005,"name":"prashkre","email":"prashkre@in.ibm.com","username":"prashkre"},"date":"2025-08-05 11:14:03.000000000","message":"Patch Set 63: Code-Review+1","accounts_in_message":[],"_revision_number":63},{"id":"2a44666bacf6b0db313944dafb2c05c15d79528e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-05 11:14:33.000000000","message":"Patch Set 63:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":63},{"id":"97ab432139df52648488d39b882c4b91d6f225b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-05 11:22:22.000000000","message":"Patch Set 63:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1494/.","accounts_in_message":[],"_revision_number":63},{"id":"0f48b2ad02cb526744d808e9830bf9269e40572c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-05 11:27:13.000000000","message":"Patch Set 63:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":63},{"id":"b5216260b19371622dcea6a6903bfb9765bcb5c9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-08-05 13:43:02.000000000","message":"Patch Set 63:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/938771/63/ : POST_FAILURE in 1h 12m 50s","accounts_in_message":[],"_revision_number":63},{"id":"cb75a771014d6e4c6cbd4d588c6b1d4efa55bfbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-07 02:05:51.000000000","message":"Patch Set 64: Patch Set 63 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":64},{"id":"3ec84fb40d5f58e73c09e7e94b49da03005d221f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 02:12:56.000000000","message":"Patch Set 64:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/112/","accounts_in_message":[],"_revision_number":64},{"id":"6a56b742d6ad121c633e495fda65ab1e658dd7c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 02:13:42.000000000","message":"Patch Set 64:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/64/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/64/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":64},{"id":"1eebbb6d440f0a42bb841144bd389a379569b45b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-07 02:53:08.000000000","message":"Patch Set 64:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/85925ca5291143eda41db39b2df44342\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f0942a1ddaf64efcad5714c14d93cc8a : NODE_FAILURE Node request 200-0000077560 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4b8c1ff62e94e74b0abcba638ef3145 : NODE_FAILURE Node request 200-0000077561 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0d213794d8e464d812cc421bb42e9d7 : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d285e31ff1c94fae813b5b03ef4bad82 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6250a699b56b4623a72ee81b94a6ed56 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":64},{"id":"83f25cbc4cf86db227b40c7641f962afaca79c7d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-07 04:07:08.000000000","message":"Patch Set 64:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-64/FC/ : UNSTABLE in 7m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-64/iSCSI/ : FAILURE in 2h 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":64},{"id":"47bb601415f4f0d69e6f5c4eaf1e0bf97fb4bda1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 04:33:49.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d71d66ec93423ca87492d9c75f4cf4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c501d6a12ed4a0c8fd2e1ffa4c71694 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd98cef728b34e9cbad3433e841dc0d3 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee2b7d1ff9834a04aecd25608606b476 : SUCCESS in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"7c97469a4dd6454cade9aa82acec9e43a2d42f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 04:40:21.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/686a4e4811864235904d9ce1b7141ff7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d001e6cb5cdc4584aeed39a3ee3eb1d6 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2df4a9e0857d46c2819761eccc40b1e1 : SUCCESS in 10m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/548dbbe1d65347219016299caa843f53 : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63d4ba0b6abd47d894462382a5ddbdcc : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cf3add6b504f59a54df89f3799e07a : SUCCESS in 17m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/2ec698e7dd8a4e1b89dcbc0105d73ec4 : SUCCESS in 59m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1120993d504e48b2806ac82aeffc2cf5 : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c86af76b1b6a41f799d31f9770f5dee6 : SUCCESS in 1h 26m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f43d4191b2324b70956a541edcde6058 : SUCCESS in 56m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3720d8751c674a03a881be7895a7117b : SUCCESS in 20m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f6090c0aa71a49faa99772b52ad063a5 : SUCCESS in 26m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63e44b0f0c7c41ce8bf51b74ea39f588 : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c654226c2a14f03bbeeb8a65b2d9c5e : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7164a50d9fb249858a8301c9d02a3faf : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d7dcb8e01cc5447f83b05bcd9e97550f : SUCCESS in 8m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d0c933be9234b46b549c553d57ae478 : FAILURE in 34m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/408a0556e85c4c028b4b3e29c13fc459 : SUCCESS in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/381954cf63274e6f88d2e14da1d8db56 : SUCCESS in 1h 35m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/43592324251748378aa6ad0cafbb09c2 : FAILURE in 1h 32m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/26c11584586a444ab6023f2ab8d6c7b7 : SUCCESS in 1h 47m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0ba976b12f0b4accbecd768f011551c3 : FAILURE in 17m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/044b3b1aa56847d18bb54828a9bf38da : SUCCESS in 29m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35f54ff93c5843ae912860bdb1fd2e20 : SUCCESS in 1h 08m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9addacf2d6904afdb8ba9c9da9170ed8 : SUCCESS in 1h 35m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a67b1714c8c746e8949f299313ca9bcb : SUCCESS in 40m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec26ef309ba042d794fcf57c57805013 : SUCCESS in 1h 39m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4a20541e8a8a475094d3a8290aa0a824 : FAILURE in 12m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4127d5d129ed43de97dec083dc0d82ea : SUCCESS in 1h 47m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d71ed7ea5b834c03a2024c3d08e8f84e : SUCCESS in 1h 47m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e58b908fc464f1b9bd4734b5a22d191 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":64},{"id":"c80b994d09b65bb9739653a2829222f46819eaf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-07 04:50:33.000000000","message":"Patch Set 64:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_08_2025_04_45_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":64},{"id":"cb8bac81ec9a442bb7cff614a2410f44f9a0bf0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-07 05:30:02.000000000","message":"Patch Set 64:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 36m 06s","accounts_in_message":[],"_revision_number":64},{"id":"a79e7f95181e460f7861c30cf4767b5eb16005f4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-07 05:38:33.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6c163c3f64864cfca200d43e7dad80aa : SUCCESS in 45m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/391971848c5340168e9740374a646f4e : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":64},{"id":"9b975a68959a47409300f2c9a845010acf883a70","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-07 05:49:32.000000000","message":"Patch Set 64:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":64},{"id":"b25811c95fc58e70e7e1bd80b5a2518ffbb85da3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 06:15:37.000000000","message":"Patch Set 64:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1515/.","accounts_in_message":[],"_revision_number":64},{"id":"87caaf01fc4a0cdbb1acc54efb8e6e582166c545","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-07 06:21:46.000000000","message":"Patch Set 64:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 05s","accounts_in_message":[],"_revision_number":64},{"id":"04ff7dda0a6fb69a265a1f212bf80c9fca6f09c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 06:26:07.000000000","message":"Patch Set 64:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1380/.","accounts_in_message":[],"_revision_number":64},{"id":"a5818863508f5fb80ebfd798da37556c11edf1b5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-08-07 07:02:23.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/63/20250807103956/NEC-V-ISCSI : SUCCESS in 5h 17m 40s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/63/20250807103956/NEC-V-FC : SUCCESS in 5h 22m 23s","accounts_in_message":[],"_revision_number":63},{"id":"97d651ade407afac784c4b13270a08b701c34b69","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-08-09 16:21:00.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/64/20250809194028/NEC-V-FC : SUCCESS in 5h 31m 29s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/64/20250809194028/NEC-V-ISCSI : SUCCESS in 5h 40m 29s","accounts_in_message":[],"_revision_number":64},{"id":"5cbab141735c634111288225ad76ac1e79e68fbf","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-10 02:50:03.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/64/20250810025138/NEC-FC : SUCCESS in 6h 20m 16s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/71/938771/64/20250810025138/NEC-ISCSI : SUCCESS in 8h 58m 22s","accounts_in_message":[],"_revision_number":64},{"id":"8dffa317a0c9e6cc8981b4ec0c73e2deff4c43e5","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-18 14:58:41.000000000","message":"Patch Set 64: Code-Review+1","accounts_in_message":[],"_revision_number":64},{"id":"d2b4413d6ea6537f95dbe88c81f605bf47eeb22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-08-21 03:32:09.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":65},{"id":"25f5b60d2a6f97a4ec120424aa67569cb5ab1ef3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-21 03:34:21.000000000","message":"Patch Set 65:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":65},{"id":"74fdc5816e301e59730b571f798989aa32671976","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 04:17:20.000000000","message":"Patch Set 65:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/236/","accounts_in_message":[],"_revision_number":65},{"id":"802191db524331bef90a357a36665bc49b040773","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 04:32:34.000000000","message":"Patch Set 65:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/65/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/65/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":65},{"id":"7f9b96f3e288c53ec3d15dc1d63dc919017cc95b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 04:32:48.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02615c8b22814eeb9fbe7be5ea12751c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c357c9b8e1d4941bc485cff7631e535 : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97ee583925b0477a88df0a1b0c8ed161 : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03ca3c09fe314a4ea01a5c8d1acdd42e : SUCCESS in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"2993685d369f26419d0ba6be082daf3dc9759d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 06:19:54.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bfb798ac84249088bbf5168f059f9f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a442f6b875406baa550a25ec3568cc : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a22eb718c2b74e478c8640e116893e1f : SUCCESS in 13m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13e3360a0b164f65bbb97af0d1eab352 : SUCCESS in 10m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/419ada44b5cd4c30997267e9f8bd225d : FAILURE in 14m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5d80997bd3454289524da5464f515f : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a5dba4f7689e49858a063d1a23ed50f3 : SUCCESS in 1h 00m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e55041d075544c359b16458728d792bd : SUCCESS in 59m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ccbb40f9ba24c6fa5f1181e28894977 : SUCCESS in 1h 32m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87ba6b66f6a148eb9598a75b7f400e01 : SUCCESS in 54m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed565227d624c21841ccdcfe4314b08 : SUCCESS in 18m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb931e1695a040abbe4604f933ad9c1f : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5580cc694544281b4873692ed60591a : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/588c1123bc804a009cbe5b5868b689fd : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c0679a19ed040d0b76cb8d851565645 : SUCCESS in 5m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2558c066ab234ffdad1fdb07d33355cf : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/70764b77d6cc41be841fe0e19d32f011 : FAILURE in 36m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcca146397a847cc8f7d34537903dcd2 : SUCCESS in 3m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0839059384be40b5bc5dbc9adb609bb4 : SUCCESS in 1h 27m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df1212445b73495ebe174e68b1f7b5d4 : FAILURE in 2h 40m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/74efaa553eb1441aaec36dd076cde461 : SUCCESS in 2h 08m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/aec507cf6ad14784b87d6ff5d585b0c3 : FAILURE in 17m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1f3bc84b4073440fac2fbd25f0e7664d : SUCCESS in 22m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ee2f4559e5f4ab3858ff7ee5862d456 : SUCCESS in 34m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b96ce8c3c7149e3a15a45af2ad40987 : SUCCESS in 1h 15m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f11c00bf0c214eb1ae9138aa1c5b79f0 : SUCCESS in 52m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95fb5b92ac5044bc9169b0f8bd4ff987 : SUCCESS in 1h 05m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5c235d0f1e1449009086f5358273ba19 : FAILURE in 11m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e211b38929f341c49ca3e69f5cd4145d : SUCCESS in 51m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb3ed0398f524c2686fbcf6979fb070e : SUCCESS in 1h 32m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17980aa565dc40e8bbb3d440a65ab105 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":65},{"id":"b2d6941888000cdbfa34cfaa47b4f484ca2e9682","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 06:56:58.000000000","message":"Patch Set 65:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1506/.","accounts_in_message":[],"_revision_number":65},{"id":"9cbe5d9080b26213852339cb9b352e925a9b8038","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 07:41:52.000000000","message":"Patch Set 65:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1641/.","accounts_in_message":[],"_revision_number":65},{"id":"e3227a6cd9e72875a79c0f9813c1f4102a360504","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-21 07:50:03.000000000","message":"Patch Set 65:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 21s","accounts_in_message":[],"_revision_number":65},{"id":"13c9d50fecd6615614844596f4a8ec8c14d7e677","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-21 08:32:07.000000000","message":"Patch Set 65:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bc43f2635c6145ef8b9bd11d3aa11d75 : SUCCESS in 48m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4146e2b9dd6545169341cc8550ab34f6 : FAILURE in 51m 38s","accounts_in_message":[],"_revision_number":65},{"id":"68300f120478a5553c48b9a4a924d5d3a2b48c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-21 09:10:54.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":66},{"id":"0a535918ed2eb81f7190b4b44cef93f9a09be31f","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-21 09:12:42.000000000","message":"Patch Set 66:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":66},{"id":"d83bb93a525459fd22156d9f6f820c8e674974b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 09:59:29.000000000","message":"Patch Set 66:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/247/","accounts_in_message":[],"_revision_number":66},{"id":"efdb4f3c60a6069f1a89525debc0ba14b9bacbd9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 10:14:04.000000000","message":"Patch Set 66:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/66/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/66/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":66},{"id":"d2419a7d5387fa41c2400d3321479207ee997c2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 10:31:45.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbdc196817e341559c90eb3fc56821ed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0886e33161354df2b54e66c214265bd5 : FAILURE in 27m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/135b9a503ca841178b878610579fe27c : FAILURE in 28m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bfbf2731152f43e78505204e943beb59 : SUCCESS in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"c18490b0685f315a25c32dd6d9456262ebb03ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 12:22:45.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa6ee22717940eba31dfad4779ed9bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6524ae8b23a94f94bc042317e422df88 : SUCCESS in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3311724bcd94ca0bbc7984edc478cdc : SUCCESS in 14m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf00c97c7fe84a479c7dd2ce56c816f9 : SUCCESS in 13m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/185f6de42bf64273b825f9e10a8a8707 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252b12cd6239413d9d3200f27646289f : SUCCESS in 16m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/1d5c946c5b4c4b7c94d46bb025092dbd : SUCCESS in 52m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0af7d7c2197144cd8e480db83fdeb928 : SUCCESS in 53m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be2f07e7d0884636a1329c197dde4018 : SUCCESS in 1h 30m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27a36fcbd2724f2c97b11f4f0a12d026 : SUCCESS in 31m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dd89d51e86d426d912b4585e86c1e3f : SUCCESS in 16m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f7db2e18d361477d90034e0b6c1cf48f : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0ec7d51e788459ab1638b1a2a541c77 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e77b047f2cfd49be9fb26d8cdc8aa274 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d1d19ae26524d24bae721644bf72be2 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bc4c4c8fd8284f21ae83867406b20f37 : SUCCESS in 8m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99aa8ad383a2411f911157883ec87431 : FAILURE in 38m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2d93cd9abd54b1296f9564bf93c81f1 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/004fb51f07284f1dbb8460439e3bd074 : SUCCESS in 1h 20m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d2ad570791c48109431e0a756d2ed87 : FAILURE in 3h 05m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e0d15ad0aaa64d27aba582b67699243f : SUCCESS in 1h 47m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/051f042d61f7474bb18dc0df50f7f7b8 : FAILURE in 19m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b96fd2b0e0ac4799a7bb621c65ab3cfd : SUCCESS in 33m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c615e6699e54dbdb96a7057de020c88 : SUCCESS in 1h 08m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e069b0d936024d0bb9c1b4d362cbb288 : SUCCESS in 1h 35m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c7d03804ff24919835414932374df3b : SUCCESS in 23m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45259b9398b64e72b70baa306066a4c6 : SUCCESS in 1h 42m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f98db2d988b747c5b89ad9c41049455a : FAILURE in 10m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23a90e3307724fae84be7a25ad7e489c : SUCCESS in 1h 53m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/883a3a9dad344465aba4297340b00e2b : SUCCESS in 1h 35m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9db6b6d1204c4581b05f4a0b5f52c7ae : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":66},{"id":"09d270da28aad55965295a1a2613a35909e6310d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-21 13:50:21.000000000","message":"Patch Set 65:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-65/FC/ : UNSTABLE in 4h 1m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-65/iSCSI/ : FAILURE in 10h 15m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":65},{"id":"db8aa16c6b382ddad369580f6b3454c30a14bb60","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-21 16:13:37.000000000","message":"Patch Set 66:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 40m 07s","accounts_in_message":[],"_revision_number":66},{"id":"83ca74a2ce35a71d1aebabddb36c25f0ec320787","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-21 16:38:52.000000000","message":"Patch Set 66:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/91fd731153774441a4c22d25c979300c : SUCCESS in 52m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1ce91875b484b7eaaf2df5521360a19 : FAILURE in 52m 17s","accounts_in_message":[],"_revision_number":66},{"id":"20929b137c17f6f4fd757a91200b0059ddb8232f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 18:29:31.000000000","message":"Patch Set 66:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1517/.","accounts_in_message":[],"_revision_number":66},{"id":"499c4fff162d5e6c0a905cba7fd17c9a02ce77ee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 19:12:12.000000000","message":"Patch Set 66:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1652/.","accounts_in_message":[],"_revision_number":66},{"id":"5c73f4b06a7c67ca4121680c046570f41ff37c86","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-21 23:14:17.000000000","message":"Patch Set 66:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-66/FC/ : FAILURE in 9h 22m 9s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-66/iSCSI/ : FAILURE in 8h 12m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":66},{"id":"abd57e395aced6dd4e1e5581f9618768fa440568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-25 18:10:54.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":67},{"id":"44b79cc860d90120c9e77d05a388fe60db1135c0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-25 18:12:48.000000000","message":"Patch Set 67:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":67},{"id":"e68309a54af9d425ec5416f51fdcab069354246a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-25 18:20:21.000000000","message":"Patch Set 67:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/daead5b3abe849099d66d2e74122864d : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e3b98c92d0124597af4eda80216d7444 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":67},{"id":"335c25a33ac2e3e09611bdf1ec666d2f64da25ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 18:55:34.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8adea8c64c246f4b49c104850139f5e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6aa0dbc755304f8d81646fc20dccf5f6 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36134938a9264c00b7601336b38197b4 : SUCCESS in 25m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ecc31d1661fb43328add8470524099f9 : SUCCESS in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"b78112f3ff3865f0a473fa45a41a08d636173212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 20:34:08.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67531c7ff909479ab48e27024a2f9185\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c278c93103f348369881ccc2b21c464b : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0fe27cfc92a4f9d978e50df7343300d : SUCCESS in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e9d9f7ad7e8485da6338cd9964bef00 : SUCCESS in 10m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61b4d3b7e5d4438ead8c90a9fbe5f5a4 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e0a166f08b40ab9c66d3372b9446ff : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c4eedf54c2b24698b2dbc0cc3a07bed8 : SUCCESS in 49m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f294d47fd6074c578359a5b0fae819cb : SUCCESS in 1h 24m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7a5a00cb1834a76b2bca6f25fd6c6dd : SUCCESS in 1h 30m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9957469809b5499baf40172ef20ba7d3 : SUCCESS in 38m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d32afd483d4b0e89f137a1ae78bff1 : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df05376366db40969ba0733672749340 : SUCCESS in 21m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/25952494f73249b29184f4cdeeec309c : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f993c442eb5c4f539986f70464ed8ce8 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14f4c5364e244f0caa0c564630dff25f : SUCCESS in 7m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/65337068bc614d21926bfe909ea931a0 : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c21545fceb9b4931b625c9ad6f31fb9f : FAILURE in 37m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f7aa2a0f1ec4a7cbbf2cf342e738b63 : SUCCESS in 3m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11acabfeb75c4214ab0ab46bb32ad636 : SUCCESS in 1h 04m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/649be2434ca144a4ab2214c8b8ffe9ec : FAILURE in 1h 44m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb5a2ce8bbfd44a7b06006d14c8a49be : SUCCESS in 2h 01m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/638db8c95123457f8d161616ce627480 : FAILURE in 17m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e8bda663d8fc443cb63b89258e0dcd2a : SUCCESS in 32m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d404c6a3ef542df86bf01b65d131e5e : SUCCESS in 41m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/28c2b14bd5e448aca9b0510ab2356ff4 : SUCCESS in 1h 31m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8fc7f8899bf948dabad20bed27e89bbe : SUCCESS in 44m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec9bbed86c9c4c9ba113116b94789565 : POST_FAILURE in 2h 17m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1900719ca9f54f0eb374b0c69a71296d : FAILURE in 17m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/faa4f903772b434ab7d342eb44a58d8d : SUCCESS in 1h 41m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b6dac0913794a99a7d7a50872122a0f : SUCCESS in 1h 29m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37e5d68832ea4d728d05d6c5de473e2e : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":67},{"id":"8927e93c2680e1ac5fd87cbdae18ba5fdec8fb73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-25 21:43:28.000000000","message":"Patch Set 67:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/290/","accounts_in_message":[],"_revision_number":67},{"id":"8efc50292088adc397d7f1bca6d16c1904a48240","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-25 22:37:26.000000000","message":"Patch Set 67:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/67/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/67/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":67},{"id":"f04fe9b2bfda165d771a9b65853e186bfd4541b4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-25 23:28:30.000000000","message":"Patch Set 67:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1691/.","accounts_in_message":[],"_revision_number":67},{"id":"63ef9ca33c5652c8b29b2eeb490867bedcc5d5b2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-25 23:42:17.000000000","message":"Patch Set 67:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 18s","accounts_in_message":[],"_revision_number":67},{"id":"7002d1d69ff104c6806e7dd03e51d331f13bd139","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 02:18:02.000000000","message":"Patch Set 67:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1557/.","accounts_in_message":[],"_revision_number":67},{"id":"def682b6c16be463eea008d776bfabdfd985480e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 07:03:53.000000000","message":"Patch Set 67:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-67/FC/ : FAILURE in 11h 35m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-67/iSCSI/ : FAILURE in 12h 50m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":67},{"id":"b738ec2be7242435b27aac72a07802419ad4b1a9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-02 18:00:10.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"956cb0a55e7373aca6cbc350b132377c43260347","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 18:46:03.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93409a41d9d1456fa192a7ea3c8f6f1c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c36cbbb0dd9b46788b27f6dd2f2e6a7f : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33d2709a3e0943ba98dfef7a403f0f83 : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad5cac33c1c44978b002fbdcd021c81d : SUCCESS in 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"3db2ba380e9fbbcb35fd0e0ee1481ad72eac3363","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-02 20:34:37.000000000","message":"Patch Set 67:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/03057114487147dc8665cc344f27d45a : SUCCESS in 50m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/06dfaaab7ceb471e9bbdce4a4501159f : FAILURE in 1h 04m 30s","accounts_in_message":[],"_revision_number":67},{"id":"05a32cc0018509247be61c82fbacaff4e225d09a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:44:43.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738cf3927643473f968dc7311ba8eafd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f996747d74054a0fae39deeb94f81af9 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/793d1f0ae4e74ace9bf61f69fc09d7f6 : SUCCESS in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3f6ac2c40d44985b5425cdbb9709014 : SUCCESS in 12m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8d4fd12df17422996a6f6a240004bb5 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c0cf851ebe41769fa77d9da6cc5cb6 : SUCCESS in 15m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/1f3d1dff1a144206a86709bcceb43bae : SUCCESS in 55m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70dea7e49b52425b8742333183670afc : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4db51149a53b4cd6895dfbbc92aba7c2 : SUCCESS in 1h 19m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecc0c26fc071417abf375afcca5d7c27 : SUCCESS in 56m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a4cd201f7084f22a518f7f4f4cb06fe : SUCCESS in 16m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e2f145a06e0d49128319292b85b8dd71 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/140f94da06cc4e16a0bca342ffa659d2 : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1544feb7a3bd49aa83ef63e5ecf76d7c : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/587fb23ff0a942279e25d28138775ddb : SUCCESS in 4m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e8003aa40c24d75aa4234aea94d2476 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2acdd12a2aa544b095f84561480df2b3 : FAILURE in 35m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1968323c793b4a3689836b2e1330dcdf : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14feac65b586408fa2f089506dd2f47d : SUCCESS in 1h 42m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e12685c5b72474e9f8b30318fb36a21 : FAILURE in 2h 37m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f136bd052fe747e083ba615cbaa30286 : SUCCESS in 2h 01m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74281e11320c479a832b77ca8129ff37 : FAILURE in 17m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/12bc7670b1014c0781a38074112c01b9 : SUCCESS in 26m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e78d21cc966e4dd3b1c75e7021f2a7ff : SUCCESS in 51m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0407ef78f65d444490bff930ff884635 : SUCCESS in 1h 25m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a1e9130c5b714e6a8d3ca6a801ea1e93 : SUCCESS in 40m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68c2f3f86af54ba78ca467612cc54dcf : SUCCESS in 1h 26m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6b3b4a35f2cc4af8b57c4766ae305735 : FAILURE in 11m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97e427cfc6b84127846f704af96aa847 : SUCCESS in 1h 36m 24s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/724b8c594efa4afe81d37b55b4228ff9 : SUCCESS in 1h 34m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/106040f6cb8e4505aec9ef026ba416c0 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":67},{"id":"73d248e4b1627a52f9676818bc668590510cd04a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-03 05:57:21.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"66602eba17b43a46a839258822096ad3cd145333","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 06:03:52.000000000","message":"Patch Set 68:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":68},{"id":"bce2f3e3b946ccbef0344dce614f0196ea8373ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:11:02.000000000","message":"Uploaded patch set 69: New patch set was added with same tree, parent tree, and commit message as Patch Set 68.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":69},{"id":"223be9f78568501de7c1a8e0dcb1e477c47733f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 06:18:40.000000000","message":"Patch Set 69:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":69},{"id":"705dae767151c726fbf11a8766b59f36290113c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 06:52:18.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e5ce9bbc284af2b3d4fe9d99648cbc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/805f93b859ba48c781a35b19c5a468ba : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f185022fca164c8790af2701e3dd4674 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ebd4f2a80f74ae09b3d92c4b162fe99 : SUCCESS in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"17765ad05d2cc068e800d7fdd029eeed517b1a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 08:45:12.000000000","message":"Patch Set 69: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5137e736b74a4d8db5904311b5a0207a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5bf95f9eba48848048b5bb61b433dc : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f1642df2cf34faba56b10418f4ef398 : SUCCESS in 12m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/505fe75edf23476f9081a2b0623cc492 : SUCCESS in 13m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/345308fe1679477f914b5cf754184884 : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30713df7313d4341860da8c3fe0edfec : SUCCESS in 15m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fb64a116e210436f8b83b11ad25430f2 : SUCCESS in 58m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a01a975803884b488446a8ceebfffc96 : SUCCESS in 53m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/147e0fccecd6436c86fde4ea713cbb7e : SUCCESS in 1h 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf9073ef24984102ac28fe4a1cfec200 : SUCCESS in 1h 03m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffeaf1a3e4194e8e8b9dc5fc8690e49a : SUCCESS in 15m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/227235a28c3c45e19de441d05f04f4b8 : FAILURE in 26m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8ec4ecb9c8a42c39f6192ef9ce27986 : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c724e0e1c6954596b2bb2d9b3a130d66 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba313511dc5a4b608bd0883d694a0408 : SUCCESS in 6m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bd982029d3054b038d0693d855e2d9f6 : SUCCESS in 10m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1843d6efaf894569a079193347749cd6 : FAILURE in 36m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60861451ef764e8d802176e14bece4b7 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22b780ac0ee741edaaf84c934dbd582f : SUCCESS in 1h 44m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04a831e4bc5e41318dad0b75b111400e : FAILURE in 2h 17m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01d7224c9d6c402fa8fc303e5a27aa3c : SUCCESS in 1h 36m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/28f0ef67e8804c2881cc95235b48a164 : FAILURE in 12m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/88f55c7a47c5465a898781b65718118e : SUCCESS in 24m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9c01cc9393c42068c30bd4b13b6eb80 : SUCCESS in 1h 07m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/87c1c484339c40f9b0b355f162b5d798 : SUCCESS in 1h 37m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3d21383898584ddc9f0a85779e54d66d : SUCCESS in 41m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/50d4b239e4a540d7a4a2d04014dcb731 : SUCCESS in 54m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bfe5a430d4d34111992ddbdaf1e44546 : FAILURE in 11m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5381331e8bf141a89432b65348619724 : SUCCESS in 1h 23m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9449c7280c1a4b559ad0d297de165fdb : SUCCESS in 1h 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f65a8ad17abe4f97b781abe81efc1746 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":69},{"id":"baa66c376573e2fb6e70bd81abf03b2eb8cf2b03","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 09:33:14.000000000","message":"Patch Set 69:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cc6df933aa1140a2a52fc6fb043011b0 : SUCCESS in 49m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/97cba64535224ee98772223842e93b11 : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":69},{"id":"f921b401043ad006040073790970eae14d4bb01c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 09:35:09.000000000","message":"Patch Set 68:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/407/","accounts_in_message":[],"_revision_number":68},{"id":"182a60b5f668bf26146542b6caf03031a784be03","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 10:17:48.000000000","message":"Patch Set 69:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1795/.","accounts_in_message":[],"_revision_number":69},{"id":"24a1c6856dd68a78ff2ebcc3c2b05a37450e7154","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 10:29:01.000000000","message":"Patch Set 68:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/68/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/68/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":68},{"id":"3c728d86872dd6045a691cf3737f700de8af325d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-03 10:37:46.000000000","message":"Patch Set 69:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 43s","accounts_in_message":[],"_revision_number":69},{"id":"f4cd03d586e934d2cc55d15e6917b14cb9e0521d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 11:13:39.000000000","message":"Patch Set 68:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":68},{"id":"131dc10aace032665050254150fba4f951cfe35a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 12:24:28.000000000","message":"Patch Set 69:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1659/.","accounts_in_message":[],"_revision_number":69},{"id":"a22160f79eaafa580753dc0a121c7eba34669939","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 18:54:20.000000000","message":"Patch Set 69:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/417/","accounts_in_message":[],"_revision_number":69},{"id":"7c41254b8b65fa8e32918243cc7b3c477a1b6e2a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 19:21:45.000000000","message":"Patch Set 69:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":69},{"id":"7806117e5f4a936e4570a34cb573877a4d2a8d61","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 19:50:26.000000000","message":"Patch Set 69:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/69/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/69/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":69},{"id":"de23976f87ae0786aa6755b997728f2e7c01d58c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-04 17:41:34.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":70},{"id":"44275467623682ecec97c8e2d990bfcd892e2ea4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 17:48:41.000000000","message":"Patch Set 70:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":70},{"id":"d5a8526021d00932c51a48a9bb75e9000a6c4a51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 18:37:03.000000000","message":"Patch Set 70:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/450/","accounts_in_message":[],"_revision_number":70},{"id":"91cb4e459a33cfaee2d576456515f4779a0d00f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 18:54:53.000000000","message":"Patch Set 70:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/70/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/70/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":70},{"id":"067b4d4aa44d928502f9bf5ed1237b44b3567804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 20:25:50.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f421f22ad2d1452a85ca5fb734575536\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cefe8f6100294446b86783add5990106 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9492856924c24de79b371e6449c5bc8c : SUCCESS in 8m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c58e076a6cb434680e4004c916cd97d : SUCCESS in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46015277d87d454989695fe76e1c00ba : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3021aaf49b1f460fabbc60d8241a19a4 : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/24155675664c4923ba41b9aa3bfdc9c0 : SUCCESS in 36m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43ea4c621fb54a81830283d871c201ff : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e65401d3a5ac42bd9adaeda390a57e5e : SUCCESS in 1h 40m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7036a85af2ba4c06bf26f4fa1384a479 : SUCCESS in 50m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d47ffe18eb64125b87a48236d4cb31d : SUCCESS in 16m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03a21dce2c3c4ac1899375efde889ca7 : SUCCESS in 17m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/838e463bbe334ef698c37562c38d673d : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83d8dfef834d42fea7edfe6d12a1d692 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad548189bdc94a1c8a3aa55805c23d26 : SUCCESS in 6m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c014f3deb3d044199769708c17d57848 : SUCCESS in 9m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d8215f4a66c47ffa73c8d08993a0211 : FAILURE in 39m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68b7428b53724117bcbd79a101eed32a : SUCCESS in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a71dd2bdb8ad47728d55b7167170338f : POST_FAILURE in 1h 03m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/914444e602e54d9aaf070c0cb8c520a7 : FAILURE in 2h 38m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/48a050385413442c91a7b9b54ca58ffb : SUCCESS in 1h 43m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a485ebc845c24bdeac608a4b153b94eb : FAILURE in 8m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d02d7cca2b954d0fb810f5d3c5a73aa3 : SUCCESS in 21m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ed18f08d46a4e0ba5c53277abed4730 : SUCCESS in 1h 05m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5620618c4e5842ffad752fbb48c3c548 : SUCCESS in 1h 37m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b71c83e61cbd47e68eaeb520554f7ceb : SUCCESS in 28m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b200e10463f24773b93ac5106a2c92fe : SUCCESS in 50m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8ef213a356bc49a382fab115af6f4f03 : FAILURE in 23m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/744eeb778b26410481a890ed12e76e45 : SUCCESS in 1h 38m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c21f62f898f7479b963460659f1e35ab : SUCCESS in 1h 34m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa9453e6d83243efbd39ae93b6c2d78e : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":70},{"id":"f4d080099d8683602791cc59d69b60a7ec69dbcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 21:11:17.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0335ee8d28542c095bbbdb5d9ad4da0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a269d3868bd4847a6ba5bb93989fb6f : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69c678c3bb334379a2a01575485c418a : SUCCESS in 26m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af8a1ab713754960a35a3153f6e866bb : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"d0568f0ef0acab026e8089f7100d4d88e419dda3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-04 21:14:25.000000000","message":"Patch Set 70:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4b58c865a50746b885f8457e92ab67f9 : SUCCESS in 47m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0fa1adf8a2da4c9b920dac821f192b6a : FAILURE in 56m 35s","accounts_in_message":[],"_revision_number":70},{"id":"b1b6f5a44b34e9390325d868674c16de7c7788ed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-04 21:54:24.000000000","message":"Patch Set 70:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1682/.","accounts_in_message":[],"_revision_number":70},{"id":"cb6b1fef1beececa9b8b987d64b61f104f1999ea","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-04 22:01:46.000000000","message":"Patch Set 70:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":70},{"id":"acde6483ea43e8e28402c913b0ba0b0d5549eae9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-04 22:02:19.000000000","message":"Patch Set 70:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1818/.","accounts_in_message":[],"_revision_number":70},{"id":"048c8d6cffb6a96b91e700d3d445a66f21f16e4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-05 08:52:22.000000000","message":"Patch Set 70:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":70},{"id":"0dd5531967fb2212e017022cedb329c4e7459b9d","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-08 10:28:27.000000000","message":"Patch Set 70: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"3b13d58b65dde1580bb3704298b38da5c75265ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-08 10:36:04.000000000","message":"Uploaded patch set 71: Patch Set 70 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":71},{"id":"aa81ad04bacc053423b26e96415a67650e55c12d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 10:41:22.000000000","message":"Patch Set 71:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":71},{"id":"7066ac0f7e633b6d0a0dd3de36ad26b233f36c8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 11:19:02.000000000","message":"Patch Set 71:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/491/","accounts_in_message":[],"_revision_number":71},{"id":"40a8cb6a3228b50f726fc4b1d813d871dfef26a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 11:38:43.000000000","message":"Patch Set 71:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/71/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/71/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":71},{"id":"88f9527c2b12fc2ffb1c134055d7543149f9aab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 11:57:08.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd4279ca7b54dcca6dd3e9762d08000\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5c7142ba7ce4151b679286ea309bbad : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99c064f57e7f44a59b081316d7316a20 : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/068c979ed5e344df9afc83a14df5af6e : SUCCESS in 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"a2037453f171fe6d7d4282d0ec9f002f69ddc593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:51:17.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35e0a7863b646369884ae0c43750b5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb64775001043a0a25aff99d6223984 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4a75c02cf0a48ff9049b7fad4ce0f8e : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00a44f7c8aff41c3a0f2e0af2ae5a880 : SUCCESS in 11m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ecb1ab10c434bbfabb848a9c829f754 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32fd3b79d0ad4c88b8c6b9f44c0dae40 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/c5086c05853a42e8b89694a99c789078 : SUCCESS in 53m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fc2d8ac3c5349d586f5cf0c98e855f2 : SUCCESS in 34m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8228ad25f0104b098a54f07e138998bc : SUCCESS in 1h 23m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3c0f32eb4224937a6dcfab809fb2330 : SUCCESS in 1h 02m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ebb16a4b50b4443a85ebd492411d5dd : SUCCESS in 12m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4386777fc6b6416b9ffc6b9c64d8551c : SUCCESS in 20m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f5669d5dccf406c85d7936ebb61693b : SUCCESS in 3m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/283acfc115114a5fa5b8a8a5abc786f7 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa8544e6fb934649b4610d3b0155fcf8 : SUCCESS in 5m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dcb1ac9845fc48579dc5874ba30f25af : SUCCESS in 8m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/38c8073972214fad9e616eabc38eee70 : FAILURE in 19m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/925c4bb7c3d44945b00fbaa90b648441 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff491aef9c504a8097f896e7d40bed16 : SUCCESS in 1h 49m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/faea6c0fb0c543c4a1c6725dd0279008 : FAILURE in 2h 02m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6bb0250a5d254d849297c2dded230217 : SUCCESS in 1h 53m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/028060341d1143be95bbc80097340521 : FAILURE in 10m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a4450db1751455bbd239d9eee970d08 : SUCCESS in 19m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/527735fe4dcf44ff9e0036470cff617e : SUCCESS in 1h 12m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b315e02f7d44b09aba8e158fe004c19 : SUCCESS in 1h 46m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c9b4fc2097e4653bfb6fa9d22d6b7c9 : SUCCESS in 39m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a6667f9026364e04a60d6d9613abd6ee : SUCCESS in 1h 23m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/64d35942580843898545362312f8822d : FAILURE in 10m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c78c238c40c6483f95e121504d6e9e5f : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/863c2f392ed64306ac7d03fb563e1e97 : SUCCESS in 1h 15m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cc4a0a564f147718c7996a1d4a439b0 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":71},{"id":"bf310c836bfe1864ac873e2863a925d2b6d9534b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-08 13:29:02.000000000","message":"Patch Set 71:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 42m 04s","accounts_in_message":[],"_revision_number":71},{"id":"99822c2594e4a87352d9b21c479781379f7cd7ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 13:36:57.000000000","message":"Patch Set 71:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1712/.","accounts_in_message":[],"_revision_number":71},{"id":"aa42350ffb78cd24c09edf6769d9e1b8ed78eccf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-08 13:37:12.000000000","message":"Patch Set 71:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":71},{"id":"9379bb36b7374c168c8b6de7dc31e9fb1eb52180","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 13:48:45.000000000","message":"Patch Set 71:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1848/.","accounts_in_message":[],"_revision_number":71},{"id":"2c2560d730bcc3dc88d72f64d98bee8f344c8558","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-08 13:49:45.000000000","message":"Patch Set 71:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 47m 42s","accounts_in_message":[],"_revision_number":71},{"id":"8a6cb9ddc8512aeeeaf284704bdafb1b495bb3a2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-08 14:09:26.000000000","message":"Patch Set 71:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d45a1439b025407e8f9f57aae403871f : SUCCESS in 45m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c96ec054ca634d07b225f659acbed21a : FAILURE in 1h 02m 37s","accounts_in_message":[],"_revision_number":71},{"id":"f1add4e9951aa61f755072c7dba93a355e68ebb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-09 10:13:48.000000000","message":"Uploaded patch set 72: Patch Set 71 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":72},{"id":"9c1e2843e66477f070afcbe9c1bb7540cde18df7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-09 10:19:43.000000000","message":"Patch Set 72:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":72},{"id":"39b4ca4a26012be47a1084b60682ed9c3a690146","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-09 10:53:00.000000000","message":"Patch Set 72:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/512/","accounts_in_message":[],"_revision_number":72},{"id":"30f1e226d5956c6f5c85efbfc50273cc6c18a33a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-09 11:11:00.000000000","message":"Patch Set 72:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/72/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/72/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":72},{"id":"e89036efd23dda63a9538749c26935fb34777f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 12:07:23.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a5884ea9ac433e987305cc60addb6d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89f3b1d6b7d34ba2967932eca5e3f169 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f55b25b73dc041e48262833ccecd26e7 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35fba67394454b9a8203b4ce0c7669c5 : SUCCESS in 27m 36s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"4cfcd7fb2cc6ca1dfc6fd8c83a835e4a60a19cf7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-09 13:13:50.000000000","message":"Patch Set 72:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 40m 32s","accounts_in_message":[],"_revision_number":72},{"id":"e091436567b74dc284cabcd1f051cb544a632105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 13:16:52.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d319c9eab8564cc7be20e4208f5eabdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc0d2b4451e046b49449af32934dac4c : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3596b34d78d461bbca789ba9646f406 : SUCCESS in 9m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/953d9484ec494aafad1f413a32877e9b : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b42881276f434fb78f6a55f39aebbe65 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f099518cdcf741149c84c9b80ebe6285 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/0e41565869114a73a5417b2c4a520f8b : SUCCESS in 1h 01m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf945bdaae864e039a89e782364cb91d : SUCCESS in 34m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2a3dcaabaad4bc580b38567037b61f0 : SUCCESS in 1h 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5d679495c614a458f263807b278639c : SUCCESS in 51m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5395c9dbb4244371b47bbba76ae2ea20 : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5f46e86c8eb4dc29a457da1a6683a5a : FAILURE in 26m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a0c36f679b19463e82eb2c4fc85beb0f : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6a2a44c3e35466d9e3af3d9cddf7c02 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee07f79d253043c498c144955eec5aa7 : SUCCESS in 5m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c694ff3eb38b4789a92d1f99c8289cd8 : SUCCESS in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eba0d446ec6741b28bb58d8f8b62ab52 : FAILURE in 36m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3784f765c351457daa0abccaaf38e6a2 : SUCCESS in 3m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7325092557fe417a80d476410f9513ee : SUCCESS in 1h 43m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/263c83ddca1c4f03a357f8f511da94f2 : FAILURE in 2h 47m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02f1f86832cd4221a178b106e65b51e9 : SUCCESS in 1h 54m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0342c48f45294d2c98cfa0392ee01e62 : FAILURE in 11m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6e84a55f7fd44de1ae7cee82ba5eb878 : SUCCESS in 35m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/884a7312f3d74c17b38e9097a2943725 : SUCCESS in 1h 03m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/537bb5cee8ca4dc89227157ccb615dab : SUCCESS in 1h 29m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8569394060854bd3a8959077f220be00 : SUCCESS in 39m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a61ef1fc712461581d5b4fde8a947d1 : SUCCESS in 52m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bbd1fc32d4ed406f817d73e898d39d76 : FAILURE in 9m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57a212f219f54889b64f1e2fa7f873f1 : SUCCESS in 1h 40m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03a398db02354fb0a5d77e21e57e9d0a : SUCCESS in 1h 24m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de82e83558f449adad332108199d10d9 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":72},{"id":"09e7f0d10471151d03ab34121f451031d333565d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-09 13:25:58.000000000","message":"Patch Set 72:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1730/.","accounts_in_message":[],"_revision_number":72},{"id":"78ea6a776c9ddc22bb16a906d88f49317122fa40","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-09 13:37:13.000000000","message":"Patch Set 72:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 40m 09s","accounts_in_message":[],"_revision_number":72},{"id":"d4be0bcbba719e20e71b1d0dac79aee53a59df9b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-09 13:41:49.000000000","message":"Patch Set 72:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1864/.","accounts_in_message":[],"_revision_number":72},{"id":"63eca7c472324f5a71351e1569f02adf486fd15a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-09 15:05:11.000000000","message":"Patch Set 72:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/44a7763611844e2b8711e79e1653ff38 : SUCCESS in 1h 19m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f539169253f74d85bd81956b7f7d8b89 : FAILURE in 1h 16m 02s","accounts_in_message":[],"_revision_number":72},{"id":"2dffe9bbd85f04298c39f37aec3a6c33ea82d742","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-09 16:49:39.000000000","message":"Patch Set 72:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/$JOB_ID .","accounts_in_message":[],"_revision_number":72},{"id":"88afeeccee6ef22334e94bf24a6c65c0beefec48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-09-10 08:32:55.000000000","message":"Uploaded patch set 73: New patch set was added with same tree, parent tree, and commit message as Patch Set 72.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":73},{"id":"fb6a246906b92fa0251c9b7fa8a227768cb870fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 09:10:39.000000000","message":"Patch Set 73:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/523/","accounts_in_message":[],"_revision_number":73},{"id":"c956b393be7e67ebf925d47d7814bdf694cb9a63","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 09:29:17.000000000","message":"Patch Set 73:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/73/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/73/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":73},{"id":"574e2e6abc9cf0900c9ca7870f20bace1da64f53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:13:07.000000000","message":"Patch Set 73:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c2a19c926c4263ad25512b6b44c28b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d03a701920448ada883035070f18fcc : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98b2c10ee06f4247bbc5079d862bff36 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ca98350ff7c048ebabbd7203f90f0d9d : SUCCESS in 29m 33s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"4e075475b5c1643873b331b053671c97aeb366e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 11:05:54.000000000","message":"Patch Set 73:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1741/.","accounts_in_message":[],"_revision_number":73},{"id":"392c536ef051d61443f89db317693aa9faa42fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 11:20:20.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a30c98d1c546cfb1934ac71292e0ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc3755df47f43efa75ab2fc98612ab7 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a549cc20cef24fe2a1642c4b41b01d63 : SUCCESS in 11m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6286269a28c64ae68d5acacf6d62d36a : SUCCESS in 9m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26e04086a8fe4c15becc6b8045549b79 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea61f81f9ad4d57bc73b65c0c7188be : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2f880ea7f7a84a6a8956bf9497f8abbc : SUCCESS in 35m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abef138d0c194f47b91b89f21993b31d : SUCCESS in 1h 00m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd9852b4c53942ea96b41da5ad7a9f03 : SUCCESS in 1h 42m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cbb339214784033acc72944610fdb8e : SUCCESS in 51m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84dd4d3b50f4445ca1f19dfcac21c123 : SUCCESS in 17m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9b83792bc3124b4cb006b97f9ee1428a : SUCCESS in 17m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b90d7c05882e445cab5fd4d4968c9f87 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/583114399c8046de96622de31db12f44 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f7d51b2db6b8450a81525c48fba4173a : SUCCESS in 6m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e665fd652c27445dbc6819afe1071012 : SUCCESS in 11m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66c523f511054f3d9d5d70811586d7be : FAILURE in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a29b84cb1a546fb8df07455fb751b8a : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c824c829154f4d47b67a77dfd5959b88 : SUCCESS in 1h 42m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a6a12d5538a43ccbbb3cbf466319f68 : FAILURE in 2h 12m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/510f7deedb15430c8559b633881a7c61 : SUCCESS in 1h 50m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6256a2dce39645f1a961d778c66771ee : FAILURE in 24m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1ea3790af2b47ed875ded3807628376 : SUCCESS in 17m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8de0ac806cc54ff2855c35c9caee9f01 : SUCCESS in 1h 07m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2262633ec636413cb700e5c1bf41f334 : SUCCESS in 1h 34m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04e5ea592d4d4735b14b308087da5f23 : SUCCESS in 40m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30fdb3ed894c45099b413de25af7504e : SUCCESS in 1h 53m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2bae11cb0d1345689dcbb79c010a5cf6 : FAILURE in 9m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0521d0133894c53b311753ee3a0352f : SUCCESS in 1h 29m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf6a1d979ca40b7a41a594ecca48401 : SUCCESS in 1h 12m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/095619d155194e099a0051642b7351d7 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":73},{"id":"6c43c6bcb25a342b853cc66882ee27195aee3072","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-10 11:24:49.000000000","message":"Patch Set 73:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 38m 26s","accounts_in_message":[],"_revision_number":73},{"id":"adbc4827dd67cb4e06c5bdde7e03814beb07dc82","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-10 11:56:35.000000000","message":"Patch Set 73:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":73},{"id":"5a3bf9016b58f4357b10951f125cc664873ce803","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 11:57:40.000000000","message":"Patch Set 73:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1875/.","accounts_in_message":[],"_revision_number":73},{"id":"2b87a7dd93ac5d46cb4c35daa404fdd73f57c0c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-10 12:12:53.000000000","message":"Patch Set 73:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eab6d9e2751f4bc6b957a52d9743f611 : SUCCESS in 1h 39m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5cf481ee7ccf4420b4934dde9da0b987 : FAILURE in 1h 02m 35s","accounts_in_message":[],"_revision_number":73},{"id":"85834f554ac78619a501977ece51eec2b98c0e19","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 14:03:34.000000000","message":"Patch Set 73:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,nlogs at https://oscilogs.quobyte.com/logs/$JOB_ID .","accounts_in_message":[],"_revision_number":73},{"id":"796b1e64cb615d722a5e67f36496e026bbd83d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-12 06:04:49.000000000","message":"Uploaded patch set 74: Patch Set 73 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":74},{"id":"89df2cee0ce21d08900bf9e9258b154df317d00c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-12 06:09:57.000000000","message":"Patch Set 74:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":74},{"id":"977007d17cb1fcc9093dacc6e4c5a6126c771701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 06:23:48.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716566aa91544590986abc66c007535d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/806be04fe4424ff1b429ff1b587ad4a5 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a25a892b90d74ca3b31e267b73e68c57 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d5159e2cc9e487bb6ae8d6df7a4c909 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"fba937275279c9f002c5226c77904e8c0cb55a68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-12 06:42:19.000000000","message":"Patch Set 74:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/571/","accounts_in_message":[],"_revision_number":74},{"id":"637c0de22408e1c32a8efd4a9cc16ed719db16ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-12 07:01:04.000000000","message":"Patch Set 74:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/74/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/74/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":74},{"id":"692d8ea25a71eadb5cff55d2b25c1e5a0214d342","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-12 08:05:11.000000000","message":"Patch Set 74:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":74},{"id":"89e3a913bf4239dda2863f7859db3a5ef20de66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:50:30.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18d88006429241ef8e47172e5dbcf606\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6179e8751ef34c94b94bd8d91d5427d4 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38e4654be2a446f8b27108d1a6f1be27 : FAILURE in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bfa8f6cb18b4fa08edc2e5e42065a4a : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4737ffc3b869426082f1344b6851fd69 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454f082e7a3047a8be65f48273dc5e86 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ac77c19ec6ab4d73954d3ed950c8ffe3 : SUCCESS in 56m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10a0c3177e0e4966b02baf2130c128b4 : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86b31d0096df49ec9a180a521ce98aec : SUCCESS in 1h 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4421f23872449e7868c2c48f4a3e26e : SUCCESS in 1h 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5205aa3580044b5fb253702be0be67ec : SUCCESS in 17m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f6a6352baac3435b81b527b1649839e6 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b92872a292f14e8fb757137aee6bb07b : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23ca994877284675bc5c0821ae96ab8d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d051760533c4bb1859dd22648802ba0 : SUCCESS in 5m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/98dd46c7876d492aba75ee94e19dc7bb : SUCCESS in 9m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf04a93007fb4cca8341b51387aaefa7 : FAILURE in 25m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8b60725c04240439f720a6faeb1d291 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac8917107bbe4b1faa8539eef2b453ae : SUCCESS in 1h 24m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbfc097fe187409598b150fe6bcf6a8d : FAILURE in 2h 23m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9f569d63a4f4beda9db0fd74dadca80 : SUCCESS in 1h 24m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/33ef4e7d0160435b9123888c0b239da1 : FAILURE in 9m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/16bf2a0085bd43978321c38241092215 : SUCCESS in 17m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76e38394816045b68fe0bc668ebca27f : SUCCESS in 54m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d77a41f70e8d4419902c2914ebb7b9eb : SUCCESS in 1h 24m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ff5b9eeb54b4c4b81a9edc173f81cb1 : SUCCESS in 39m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/450218702bc34aeca7647947ff8b977b : SUCCESS in 1h 30m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/42f4e14ddddd4e5b9cd547dadc20678e : FAILURE in 9m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/377fbfe30a2d45d0838b874b6c6f338f : SUCCESS in 51m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/803a16d3b3ad42aba07b6375f48f950b : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":74},{"id":"acaf7f782c434c19630df03a464991360a7ce534","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-12 09:15:40.000000000","message":"Patch Set 74: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"d8c24d06931766c09ea37c2a2a99923940cab482","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-12 09:22:12.000000000","message":"Patch Set 74:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/938771 : FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":74},{"id":"c10feb33d8746d41ac1807566e4caa389f7c9f74","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 09:32:19.000000000","message":"Patch Set 74:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1916/.","accounts_in_message":[],"_revision_number":74},{"id":"a4f7ada8cb2868da179f8344290d7c15ba4d36a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-12 10:01:00.000000000","message":"Patch Set 74:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 25s","accounts_in_message":[],"_revision_number":74},{"id":"d3c806cdec3ac4d18d489864255c36b0b7a8a32c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 11:25:04.000000000","message":"Patch Set 74:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1782/.","accounts_in_message":[],"_revision_number":74},{"id":"b185598dda7cdb8d68af88907346f0023b5fb3ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 12:03:35.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/048b5fea9a4e4e51819a10b66389e4b1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fabb499bdd194dc9b088379968078223 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a879e5a143849d2b68cc69c4151b08a : SUCCESS in 28m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a908fe47439413ea65f5940bde3cfab : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"a1b6a529555a7964da9e43f194a46dfbf4bd08fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 12:11:29.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f923aba11a1e416ead9f0b6d032f7651\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f25b85dd8b344b2bedd237e1cf1ef60 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3bbb8c22e10428a8b8a8bceb6ebc7b8 : SUCCESS in 9m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb6d3a189df34bbca7ba7b1971a85c33 : SUCCESS in 9m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b6ad70bb00f441097dd5975d31f759d : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4683feda3dde4490899f6ad536c32ec7 : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/f1eb9d5d226544b2926fe40a390b5e7b : SUCCESS in 46m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e4c7afaccc2463fa31f7099f01597a1 : SUCCESS in 39m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8646f3b7d2549f6b716fb43c5025ceb : SUCCESS in 1h 27m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4e084854ab641df8b0048413371a1f9 : SUCCESS in 59m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e06e63f64d479b81b44c1aa6578ebb : SUCCESS in 9m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f1931d4fef24d3f958c272370203ec8 : SUCCESS in 23m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ad2595d15ca140069039466492278891 : SUCCESS in 3m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a69c7b2498f44ea5a50f063a646bf837 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/506d109ab1a84cc485446dda26ed793b : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c2717478f79e4030a71db9b5380fd670 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02e37e95a1f747e786265a2d6de4dd2f : FAILURE in 23m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a0552c01a3b447684d36ff857f3817f : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/977b368423ce42cbab75d9de6d817baf : SUCCESS in 1h 48m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/596645b181b84bd9b03403b343e0a0ac : FAILURE in 2h 50m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/058b455826c14c02aec24a63ddac7955 : SUCCESS in 1h 37m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ad271545454c4156a500446ffb8f2961 : FAILURE in 17m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8afcfa479a49416b81047e42da555496 : SUCCESS in 34m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00a4d44dda7947dfb05f8894563a1a3e : SUCCESS in 41m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32bed56b07ee42c9895ca7aa4d4fa83d : SUCCESS in 1h 22m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7698d45d577f4f7fbec39234d9e0cd47 : SUCCESS in 40m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/636bf18ee25e4d40abbd6c3b26483b0b : SUCCESS in 1h 33m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ace95f9ff00a4abf83dbdfc2a6fe33d5 : FAILURE in 13m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9bc17ff794f4596bb66a5dd6fcaad8c : SUCCESS in 1h 33m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6a9186a5e09405fb36bef48b70d9fcf : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":74},{"id":"96501e0d94beb842eedc816f56def945a71a1d63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-15 04:32:38.000000000","message":"Patch Set 74:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa80b466b312415a833d6d3c30a5a0a1 : SUCCESS in 44m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/43b5d283c0ab4bb785e1d79697161f15 : FAILURE in 50m 26s","accounts_in_message":[],"_revision_number":74},{"id":"d35fe1fe63282ec611346be73636f720532d7263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-10-09 19:06:48.000000000","message":"Uploaded patch set 75: Patch Set 74 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":75},{"id":"faab7ace25934288d6cff048bdb0af1d9dbd534e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:32:23.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d101ba6dc8bb49fc9884e6b2098dc8a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b590eb517b7b41109f2df3a0c2b26493 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/016a5d6b1f0e4df3b6d0cdb9a7156562 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32b9196875d440709e33c460483a3114 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"660ae2820afe97cf4a67739501bd0fa7b95999c1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 21:57:25.000000000","message":"Patch Set 75:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":75},{"id":"f784129b057e85acd16a0299569ac8511d8077f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 22:01:50.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff8474d6d81d40ea818d561fc4989064\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d9fbc961094c9db5e1cac889a24607 : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/037db3a1de224b4d921950647c519da8 : SUCCESS in 13m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b2c7d31ad184f21aa79ff2ebb405f10 : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d07201f1caba4fa4a8d581555b38f70d : TIMED_OUT in 41m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96bcd553fdfe4877b218c0edb98f06bd : SUCCESS in 15m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/3797a0dc99134a0793475e7fdbafb487 : SUCCESS in 53m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05cc47d9af8c42dd8fe7d5eb5e00e2d2 : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44d7c4f9626c46418f934737c2c2b4e4 : SUCCESS in 1h 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e8253c60f0042ea9633f25c32145825 : SUCCESS in 56m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7591a8299f2a4e93a8a76dbc1e6a44f5 : SUCCESS in 9m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a79f1b4b7524452db81d9fbf56b5b995 : SUCCESS in 16m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4ed94cc3f90b485dbc976e09846cdd25 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cdfb6709be2a4dadac8f521c7668a350 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8716dafcb964547a1cbcdc6478b9ade : SUCCESS in 7m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59181288ac724d68a53ad16de2ab85cf : SUCCESS in 9m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e4ee7d058ae24e6782412d4d63c1b7ff : FAILURE in 41m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1a5537cd16747b19d95778c255e20a3 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bccc7a3248fc470086b522592b9c479a : SUCCESS in 1h 56m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c0621bfdeb654ee7964c78e5e9f8844c : FAILURE in 2h 48m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3bbcc2c20fd3483396987c181d86757f : SUCCESS in 2h 12m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/90a10d8cb2cd42b89589d946c79c604b : SUCCESS in 1h 51m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/69a5cd6a40a34deb80b377ada291e118 : SUCCESS in 30m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0fdaaa82168345adb91659f7a37ee49e : SUCCESS in 1h 08m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6b1ec97ceb64d4ea25ddb782bd0a18c : SUCCESS in 1h 37m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5540e5de8d3f47fba990053a0bfa0a8b : SUCCESS in 46m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0a24903042c410292a587fad5cc4e24 : POST_FAILURE in 1h 55m 02s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e3973f624ac349a291c5e6941dcb2c7a : POST_FAILURE in 1h 28m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1594d7acddf44129352d7100d6484e5 : SUCCESS in 54m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf3fb2fd4abb4ed7aab125162e39ac0a : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":75},{"id":"c61bcf0e7e520f16d05e815c10f2097105a92347","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-10 08:13:12.000000000","message":"Patch Set 75:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":75},{"id":"1ac9f96690ce1fb3822568a204ef2936379804b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-10 08:19:18.000000000","message":"Patch Set 75:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5b61caad6cf34c0ea0bf800a1c30e7fa : SUCCESS in 48m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/24f4528f156d47d5b5088d5c4894d77f : FAILURE in 53m 28s","accounts_in_message":[],"_revision_number":75},{"id":"16d4854571c0a79e02c81457048698b8b39a1fea","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-10 10:38:28.000000000","message":"Patch Set 75:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 25s","accounts_in_message":[],"_revision_number":75},{"id":"da20dd72be2e7b844cbe4c46c1c7615895679fa2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 08:08:42.000000000","message":"Patch Set 75:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1989/.","accounts_in_message":[],"_revision_number":75},{"id":"8b765aa3e410e69cb44e09bc4c51ff77b9328f81","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 08:35:17.000000000","message":"Patch Set 75:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2103/.","accounts_in_message":[],"_revision_number":75},{"id":"10253048988dbba63831c60ec7d4bc1ddf570165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-10-16 07:08:57.000000000","message":"Uploaded patch set 76: Patch Set 75 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":76},{"id":"8808aedf605275a4e7f53c68dd484cdcb8185a59","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-16 07:11:12.000000000","message":"Patch Set 76:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":76},{"id":"8a4ed235c5c67aac9cf9f3753a0e62b6c614ca7d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-16 07:15:01.000000000","message":"Patch Set 76:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ca9c0ba1865c40b6a60a1b6a912ee5fe : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dd055d56e6e845cf8a89d915ed5c7b39 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":76},{"id":"ecfa2562c5ba9ee8a16e12cd42384ca4f461fd53","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-16 08:08:48.000000000","message":"Patch Set 76:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":76},{"id":"a0251c711b1bede1c7dae0558ffa21927c8b29b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 08:12:22.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce0d8f7006c54d39932919ce201e8df4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff92cb0f8bad434dbe13dc855f94aa01 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3095f327c6ad46b9a9d8d786fc2f999c : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4bba5130502c4de3b420abdd07ad25f4 : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"5cc2c077bfb7a6346f126a1cc30ef9f1de89fab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 09:59:25.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb70cfd9fdfe4cada33bd7ad47a493ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d0dc15d5c749eab7e661820c7ccc50 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f90d2eb985a7486abccae3af0bc2e8e9 : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1a3b2c0fa5e4ba4afe9d528374a1ec6 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c4094824364058954d929756b255b1 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/3563b44bb9be4beeb15e5fbfd9a3e7f8 : SUCCESS in 54m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba2ba9c4fea34f79be33ed84a02a58a8 : SUCCESS in 55m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23a7a48344574c9fb79a90998d98bbc4 : SUCCESS in 1h 21m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c0238c570b143efad40bdf1f3df4a07 : SUCCESS in 53m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adfa3826ed7243798b2a3b79e68cf71a : SUCCESS in 15m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b55a73b752364bb0b47616aa97a68bf5 : FAILURE in 23m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2db1381086a44b8988be2a1fca6b9c18 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5134bf8c933441369749d60210680d67 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f72a4b4475174a039438e75b8c760980 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e00813e2a5b4ea09f39e3a3670a5d78 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c5d199dd7054c3d83ba426f6d12f3fe : FAILURE in 29m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62bf055782e54526a829734f7a05bff1 : SUCCESS in 4m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/219ce195ee1841e792fd71f4e77c5b82 : SUCCESS in 2h 17m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a01d3e8036d4f25be98e1395d0cc415 : FAILURE in 2h 39m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d73d9d7ac06646a5ae3e00d6a9d534a6 : SUCCESS in 2h 01m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4a508d0d66db47be80f195d04edf162c : SUCCESS in 1h 48m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/885de000dc9f440cbed985447e1561eb : SUCCESS in 26m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc6c3ae1219040eb853d5bd567530e43 : SUCCESS in 1h 14m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8cf1e6b236540b9b6f399c30c7a3f69 : SUCCESS in 1h 30m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c12fd24faef4b5faff2c72e15ce20ce : SUCCESS in 26m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77536da230ba4b32a09794859bcea4c7 : SUCCESS in 1h 06m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e53e5d41ef664846ae4147f7db4983f0 : SUCCESS in 1h 40m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db4f77c8d60a4c018fa443c5f5adb61e : SUCCESS in 1h 24m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b5faa6cf2bd4d1c9268fe460d370650 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":76},{"id":"632c8a1885a178b916683167bffd0fd605cb3d6f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-16 10:05:47.000000000","message":"Patch Set 76:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_10_2025_09_59_37\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":76},{"id":"4b4a72c547e161bae716ddae118d9b75c4388786","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-16 10:09:31.000000000","message":"Patch Set 76:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2158/.","accounts_in_message":[],"_revision_number":76},{"id":"125a0e075c262b5c30c0ad8e99169eaf422361c0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-16 10:17:12.000000000","message":"Patch Set 76:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":76},{"id":"5aede3f0d903a3c7defbd7c05d33f545c89acc4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-10-17 07:18:43.000000000","message":"Uploaded patch set 77: New patch set was added with same tree, parent tree, and commit message as Patch Set 76.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":77},{"id":"77eb595453d12a4d0d2d6310aec05b3c0fb0310f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-17 07:21:00.000000000","message":"Patch Set 77:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":77},{"id":"3826789029a0c93b71818833f2f33ee4447e6bbc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-17 07:26:42.000000000","message":"Patch Set 77:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/94d4b88c2e904c8bb0613f8963e30d5c : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/df3306afeac64df993e4611e095a9b21 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":77},{"id":"36d7f085d616591e4fee4044bbf5bc06d663e77c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 08:00:46.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfdb861f20e84b788f84715fc93bb280\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7de5ec785994d1b8b7b38684f6bf832 : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab9bd02f5f1c45148ad6c9e4d2314425 : FAILURE in 31m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/04c0efb0963e400c9a5379abc0121138 : SUCCESS in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"12610b22b270f34409851b506e0f7b30e201d038","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-17 08:18:37.000000000","message":"Patch Set 77:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":77},{"id":"80d38aa7bc62939c8ccee15dbc5aef98d5a14a0a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-17 10:23:22.000000000","message":"Patch Set 77:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2176/.","accounts_in_message":[],"_revision_number":77},{"id":"02b3a9c3a1177cb1faf2b0020202bd165881dd91","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-17 10:24:38.000000000","message":"Patch Set 77:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 25s","accounts_in_message":[],"_revision_number":77},{"id":"47127377109a1f9594b131974e2d53dfcbeebd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:34:03.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93496ae7f6da46caa8167b5e59d43147\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98f03c9097b428eb5612197d756e0f5 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/734a59bb7c0245d3af9c79386ec5139c : SUCCESS in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05e2978d32f5409288d3dcd4a8f1d127 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3a2cb03b3442caa147f0842d2a6264 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/91a2c9bd0bec45aaae01c1cc1a166d24 : SUCCESS in 29m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1277205ff02045cb957f1d5b0c8b0a24 : SUCCESS in 1h 11m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ee86dae3a0742e5a5db9d84b4d61857 : SUCCESS in 1h 26m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e960d9a5a344369b0507655e0c7dd98 : SUCCESS in 1h 04m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e9b0180a434af0badf989949b0f8b1 : SUCCESS in 16m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29bfe10c1717473db670fba027915be5 : FAILURE in 23m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/56000bceee50407fa8c553328d80f8f6 : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94005380e00f4516b5f7803fd02f660b : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8101487aa0374c789cabc550400b85bf : SUCCESS in 6m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/de0b7386bd98410cb402b5cd40ecb653 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4836193f6f145e1af15967c0d700330 : FAILURE in 30m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fb20c61b7674d0f9e12837a7857ce50 : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f69366aa3f0c450b928c872fdc63ed45 : SUCCESS in 1h 43m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a51d7d2376794e9f9df0e7061a1bdedd : TIMED_OUT in 3h 04m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e8a1a3f69e841aaa9bb4bd8783c9a6e : SUCCESS in 2h 02m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/26c6083ddd8b4a62a222ef645fc3506b : SUCCESS in 2h 12m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b831bc6c00bf4044b3e6d702c2e859f3 : SUCCESS in 18m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2fd615315d1a4726aaada7189c0ba929 : SUCCESS in 1h 04m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/530bd097f4004fdcaefc0c21a6a08f12 : SUCCESS in 1h 31m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f3667bc5782c4bdda6eebeffa096d1e2 : SUCCESS in 36m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d30e0c1f14274f49a7a66c4339ca9ef1 : SUCCESS in 1h 11m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6771ad7fd32d483e98ac93fa1f899d5b : POST_FAILURE in 1h 34m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71d6bda46bc646469de3c3d4db946ea2 : SUCCESS in 1h 51m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb7aec3d0ddd417084eb7c185d497897 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":77},{"id":"2acfb34c40c297a03cbd730077232605295e325b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-17 10:47:26.000000000","message":"Patch Set 77:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_10_2025_10_40_51\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":77},{"id":"e3909b893ac66d8fa83878e884f496f2c2f76c61","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-17 15:45:42.000000000","message":"Patch Set 77:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-77-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-77-iscsi: SUCCESS\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":77},{"id":"4ca63b633eaf66ad99617cf7e05eb1d19641af24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-18 09:38:36.000000000","message":"Patch Set 77:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2062/.","accounts_in_message":[],"_revision_number":77},{"id":"7a9fa7083ff4dd05c7779417585f7ae93681bf7e","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-31 00:14:50.000000000","message":"Patch Set 77: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":77},{"id":"c0533ab697f58bb9a8e33dff9bbd2ba1fd1b8fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-10-31 06:57:42.000000000","message":"Uploaded patch set 78: Patch Set 77 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":78},{"id":"073cf743d1071cc70cab486d9fc8aa8d2437165d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 07:00:17.000000000","message":"Patch Set 78:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":78},{"id":"f34d1c7d1ac9e224958bdfefc51efd322fe7122d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-10-31 07:07:32.000000000","message":"Uploaded patch set 79: Patch Set 78 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":79},{"id":"a8eb446f6b773de9d116e5e4c608dac482253b7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 07:13:14.000000000","message":"Patch Set 79:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":79},{"id":"811349d8c9ed4c638563a266d90b601f438f4ce9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 07:43:50.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c59e7145b54a0d97df0a2bc769ce9c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab2fb927b15546c385541e2dbb068687 : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b08db832f2104724a48d6a9de7819049 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2318952290d44f64b422f655b1a2c251 : SUCCESS in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"f49400eab0135e011f16106206f6685fd9e152d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 08:01:00.000000000","message":"Patch Set 78:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":78},{"id":"ca07af6ad9b0f0e9ffb9c17e29e79ef326747627","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 09:41:27.000000000","message":"Patch Set 79:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":79},{"id":"edfbc66dd3dc552047ad5ebff9083ddecb94c88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 09:52:40.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e5f08626e84cdaa1a036cdad785919\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dba733e502b4a2b9d16ab7a37bf5a96 : SUCCESS in 7m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d01ed3b41664ee7842f928d0ea4b5ce : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/497ec1807c7c4990869b4695eb873775 : SUCCESS in 17m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8dfde3af2848bab21c02bfab553ddb : SUCCESS in 17m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/fb9dcfab73ff402aa8d571d87e483840 : SUCCESS in 34m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9383d592136d4f789a98804ed4ddf4f5 : SUCCESS in 59m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c6fc1f591104dc7bd2513ff4777ae47 : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c27aa50770475c8f79f7ba8c19beb2 : SUCCESS in 1h 02m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72bf722725145268e42674ee200af32 : SUCCESS in 16m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c63ed4181f814a86aebe296034536172 : SUCCESS in 24m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf39ffadf5494ba8bbf3a9f5ac2a2c91 : SUCCESS in 8m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8cb3b416ec64de7b84c15603f55c141 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2dd53af8e764d4ab3de7935ed25f57c : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9dfff6582deb4ab9831232fe5a6313bf : SUCCESS in 8m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1afa78b36b14477d8d521905e2255132 : FAILURE in 39m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7eef04a20f5542ff95f8200da31aa9bd : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4aaba0c552274ddaa93964009506824f : SUCCESS in 2h 27m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3beb76c85918443f9500efc87bffbae1 : FAILURE in 2h 15m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/386a4ac7c2274133948a0bb9ae4c5891 : SUCCESS in 2h 21m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7394d4da281043dcbc5266f89bedd2f7 : SUCCESS in 2h 06m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/802333d2a61944e796d1d38ca5bc8e93 : SUCCESS in 34m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b0543280b91d4c3bad72900d50c9c16c : SUCCESS in 1h 12m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ccdd229b31cd4c66aadb336c52e79946 : SUCCESS in 1h 52m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f6b902149b547f5a089b61899f3af0b : SUCCESS in 36m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b8f883c546f40069873b42453388a83 : SUCCESS in 1h 55m 41s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0b7e0e3ef7fa4533b1e230d7d43bf43a : SUCCESS in 2h 00m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad1a632fcb174bea8a8dc76a1c3183c9 : SUCCESS in 1h 40m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0cab89a2b14426ca12d8e46f71091ac : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":79},{"id":"cf7d30358e1b801866d0a3a0bb46182e537a241d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 09:59:22.000000000","message":"Patch Set 79:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_09_52_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":79},{"id":"82bfbf1876eca87239399f10232e3bd1a5a7675d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-31 10:19:57.000000000","message":"Patch Set 79:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 10s","accounts_in_message":[],"_revision_number":79},{"id":"5a965fd6402060610095303465e884468cc5bbcf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-31 12:57:10.000000000","message":"Patch Set 79:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/16a1a3573eeb43888c666bd63ad4295f : SUCCESS in 1h 34m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3abb73c6ea854f61b4a0a7ee4f70d0ef : FAILURE in 1h 46m 12s","accounts_in_message":[],"_revision_number":79},{"id":"6839de49e2284b380ed42e59a7081320b04b89ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-10-31 15:32:34.000000000","message":"Uploaded patch set 80: Patch Set 79 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":80},{"id":"06e1deeef50d77b671095c2167a626a303f65fec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 15:35:33.000000000","message":"Patch Set 80:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":80},{"id":"7d23823a330700a54e5fca240d30a40d9c22cd15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:36:11.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff7ad2f0e1c4345a9556fb15a6a6a4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/351a617e854c4f7da6b4de5efcc861bd : SUCCESS in 23m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac58f12039354ad0a350f88cb7541a61 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5af0d692d70d47e3858eb925aff7559e : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"7a2df0961245fe1c0473099c970de1f3bc0c66d0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 16:41:11.000000000","message":"Patch Set 80:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":80},{"id":"68312b8c2905a0125dd07f3ed39f317c470915dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 17:53:24.000000000","message":"Patch Set 80: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275375ea4a6a422385a922ed433d7682\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a94d66144d24b78b1af386c254c34b6 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07ae1b3bd362433f86427805b49298c0 : SUCCESS in 12m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/724093a65fed4da3ae9ea9684f005ffa : SUCCESS in 14m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0462e599694c2192961ccc05156969 : SUCCESS in 15m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b7e3bc2b85ed4e1a950977fd12fa2bb9 : SUCCESS in 1h 04m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e60e3373c56a41e28589a443d3290b70 : SUCCESS in 57m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b71dfc8c04e4c6898e8e60b638da432 : SUCCESS in 1h 22m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f77e8b679ebc49cdab0b724e5360743b : SUCCESS in 1h 06m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ad0fe3503a14ca1a182ef583fd368bc : SUCCESS in 17m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52aa5b3e32474dc5952045a8f84928eb : SUCCESS in 31m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3c8a1ed833654bb2a85718eb0168cbe4 : SUCCESS in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/33afc10244b749ae807893ee7bdb34c6 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb2195e392c94c01b4f8076fd2e97b0a : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d542340e38444df18b52792a8530c93c : SUCCESS in 9m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c22f7be154324bff94710b8c4f7a177c : FAILURE in 36m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c139d68d0cb4f8a8b9575bda9d51ae1 : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/637058ba30f343df9cd6910337d4d1b8 : SUCCESS in 1h 54m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e624c12e0afd4a748dcba7b4c55be4e6 : FAILURE in 1h 49m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a271a1cf7754d8c9c731a7b4681fe0d : SUCCESS in 1h 45m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3aa0ceaff12c4b10923078912bb2db1a : SUCCESS in 2h 12m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/734ee1382c4648ed833154ce6fb7fce4 : SUCCESS in 33m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45b75f39d4f54199b01df123379eebad : SUCCESS in 1h 15m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/613820af1d614334af47b81944af26e3 : SUCCESS in 1h 37m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ffc07355b0da4460b36a4eb942a54a4f : SUCCESS in 44m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f7114a7df4884dc28ed6c88aa5fd4dfe : SUCCESS in 2h 13m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/cdcdeffad1eb41b8910dcf9200edd756 : SUCCESS in 1h 20m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/391105a301e04501992632a9b868e4d8 : SUCCESS in 1h 59m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/983b02104ad24c28a91a3bd1f93f7bae : SUCCESS in 54m 24s","accounts_in_message":[],"_revision_number":80},{"id":"36baada45c920080c199db370aa348566960dbdd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 18:00:07.000000000","message":"Patch Set 80:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_17_53_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":80},{"id":"5b1beffa51866491b25a78b5755d66afa9c7ac44","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-31 18:39:00.000000000","message":"Patch Set 80:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2362/.","accounts_in_message":[],"_revision_number":80},{"id":"51f28653f2f2f67355d0c92d5bc1d6f951627ab1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-31 18:39:13.000000000","message":"Patch Set 80:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2248/.","accounts_in_message":[],"_revision_number":80},{"id":"96ef9891e5087be0ded956f70c01a5f25ee17d56","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-31 22:02:07.000000000","message":"Patch Set 80:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":80},{"id":"3bfd2fa6395b0ffc3a7b5c94c3bd8fe50b9e7e13","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-01 06:29:53.000000000","message":"Patch Set 80:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4690b011677641ca8ab34799731066ad : SUCCESS in 49m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c6145be49cc14458ab67e5bd6432b56a : FAILURE in 54m 25s","accounts_in_message":[],"_revision_number":80},{"id":"b767011fc888b1a6855aeb45e3056f60e43dab87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-11 02:17:09.000000000","message":"Patch Set 81: Patch Set 80 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":81},{"id":"9121dd305e2c693009fa00f23a0e34bfe9598bdd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 02:19:07.000000000","message":"Patch Set 81:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":81},{"id":"668701f0418afe6c37168f20bff7e26463ad7600","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 03:18:53.000000000","message":"Patch Set 81:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":81},{"id":"293c79fd491b77f367d39c4ac6cb7d8f388bb1c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 03:44:14.000000000","message":"Patch Set 81:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35cc1a065e704bec96cbd44d2acc60dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/201ea3b32e214caba0aa7693a79fd0e2 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/acd12341224e461486a82114735146d3 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7e6cf17537242faa4aee8789cb17006 : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"90e207af86cfba46ffa2249ac14e9820c17d6f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 05:10:07.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf37b27aaf64359add2e65c2cccb7ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf325bb46c6c417ebd8285f29c7b730a : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20702e0f69b3441ca255bfe5026fd9df : SUCCESS in 10m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d3d88bb9a5447f9bfadba74ba09d70b : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b0a7f21f3c4ed49c87f0e5732570db : SUCCESS in 15m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/19822949b19940c68886dabe571b76ba : SUCCESS in 54m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65745a99ea434aa3b5c0404b9b51631d : SUCCESS in 1h 00m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81d26b909d984a02809f6db61a0f8b74 : SUCCESS in 1h 55m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83fd22fd0b614b25b6e5ad38fb30cc46 : SUCCESS in 57m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/334e6b47498f4d128bc6e4d6e560a295 : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/634e7adb8b0645d7a4c62e3935ec06ea : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8aa581bb82984884b86d29b5fd129348 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0783866c09b94e17a314f8579f84804d : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a8b01f2de3c4a9185128ea1efe62e80 : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa469f9af2b74c9d835f496c4ae2bbc2 : SUCCESS in 10m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bfa25198dbf241e78f43fa83137ba932 : FAILURE in 34m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e62de92bb474ac88171a2230bb102b1 : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8ab0eb9f4d24ae0aa85ba6f805d59c5 : SUCCESS in 1h 51m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/838e2da3752041e296fd070e9b73e627 : FAILURE in 2h 26m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/007d1eaac7be4c95ae94657d902ae6d7 : SUCCESS in 1h 47m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2120b4094241450db6262e0e0af8dd2f : SUCCESS in 2h 09m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aef0befc02db4734b7edb7382c7b073d : SUCCESS in 36m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6219250146c42a08b19aeedca6a0130 : SUCCESS in 1h 05m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17e073f5e32d4a4f9f3b6ea6aaf010ba : FAILURE in 1h 46m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ecd5f3ae90a4971a94aa45aa27f2e6e : SUCCESS in 49m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a93e555d54454e08980c3bd487d05abc : SUCCESS in 1h 56m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/572c21e2d01c4444888708e1c19ebe7f : FAILURE in 2h 12m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a66c0e0f62e54a829a06e60a25c434ce : SUCCESS in 1h 22m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3040b9897340ef8f533b446471e4ba : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":81},{"id":"91e79efddfaa29499ba931fdad9f1530292b3346","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-11 06:30:11.000000000","message":"Patch Set 81:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4f6c69f3bac04b9197874a944b0b631b : SUCCESS in 54m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8e226a8feb0c4071b6c9f1ec542c70d2 : FAILURE in 58m 25s","accounts_in_message":[],"_revision_number":81},{"id":"d695c2987af929c39835a5f1a703a0f1fe481841","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-11 07:16:39.000000000","message":"Patch Set 81:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":81},{"id":"3b98e937d6ff7eae6ad0a3881a6af4accd67fdf9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 07:26:41.000000000","message":"Patch Set 81:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2485/.","accounts_in_message":[],"_revision_number":81},{"id":"92868b55c6c7517ab8ffba896499a328f417e545","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 08:48:52.000000000","message":"Patch Set 81:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2371/.","accounts_in_message":[],"_revision_number":81},{"id":"8ec27976321b24f42c2805a2d5e3b48b636e90db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-11-11 09:10:30.000000000","message":"Uploaded patch set 82: New patch set was added with same tree, parent tree, and commit message as Patch Set 81.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":82},{"id":"d1026e764066c3c9deef960a7867b71b3adb2dce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 09:12:47.000000000","message":"Patch Set 82:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":82},{"id":"5b07b80f70b8dcb30759fbb86db343ddb22ce101","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 09:36:48.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7940b3b2594b4393e2744aca792c2f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eafbf6acb2f54d7086b63eaed50f29ac : SUCCESS in 22m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd448d64efca4e8397ef121a5cdb1a6c : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41e35aeb78c34f138c2edcb7c1923e20 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"67b45b961d4a2acd23620b3fc4f3ec5eb24b17e9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 10:10:00.000000000","message":"Patch Set 82:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":82},{"id":"1d2bf62c574b330333913ea39fe315b892c282cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-11-11 10:20:43.000000000","message":"Uploaded patch set 83: New patch set was added with same tree, parent tree, and commit message as Patch Set 82.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":83},{"id":"17b222774d489a14d1f132815e1daf08d0ccf48c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 10:23:16.000000000","message":"Patch Set 83:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":83},{"id":"0ee4a0e3946880dd5e046e6325c08fb163c464fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 10:52:00.000000000","message":"Patch Set 83:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd52b91a73484017ba80de780d70af5e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc221fb0446f45a3a90e011a0b23cd8e : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2b324a9bd89438699622bf9fb0b3c04 : SUCCESS in 28m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2d60b12be5441ccb862d7ac80155b60 : SUCCESS in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"6da0ab4e5e8f7c4c1858d15882df5c6b444fa072","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 11:22:01.000000000","message":"Patch Set 83:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":83},{"id":"0057e51a62a5f0b10fb862c19049e554da02b9f5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 12:32:04.000000000","message":"Patch Set 82:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2496/.","accounts_in_message":[],"_revision_number":82},{"id":"0f7ba63064a5d19bcd68d4431c9b0afdeb265b66","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 12:59:00.000000000","message":"Patch Set 82:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2382/.","accounts_in_message":[],"_revision_number":82},{"id":"d10b506081c02457b1690b0c920ca170c0101891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 13:08:17.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380d713d72864c7daef531055336e020\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d685a80b63a469988746847dffaf35d : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c57812e2dad45bfbc0f08389e356b39 : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50eb154dc9014db7a7898bf2a13e7654 : SUCCESS in 14m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5c28e0be63478c81a0ee34dfc951ed : SUCCESS in 15m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/91b8dd71f97e4176bf25ac3e7c701367 : SUCCESS in 57m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d280b254bab4253a8416598c5ec4f6d : SUCCESS in 39m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1a9ccb8c4004bda87cc2dcea616e907 : SUCCESS in 2h 00m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f8b9ff5d210445aa470e7d78f312d92 : SUCCESS in 55m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cfbbcd347bd4c42b9d98263f2b2d3e1 : SUCCESS in 16m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/070b58fc47f54349a04e7da9a5cffa16 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c91a8ba39104ea685da7a42153bd057 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b93d7442e4ae44cf92e5f398d9027c51 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c083ba808b2644fa82b114f757239e70 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/96438e530b864aa3852be499f597b082 : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e785cf47644d4b6cb2800dc31f008165 : FAILURE in 44m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/28c4e60c16174a3f88556732a90ba5a7 : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/092318f207154928921d001c5a7c41f9 : SUCCESS in 1h 34m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/13f68690a6fc4c489f7e560e3d85fe4f : FAILURE in 2h 41m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b48f335dbd5480c84232e6434d5fd8c : SUCCESS in 1h 57m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f3c055cc48414883ac491d7158b36f29 : SUCCESS in 1h 50m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1075794a4eb42678adbc4224ff8355c : SUCCESS in 28m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e7ae85026fe4a558831a6bedafb319d : SUCCESS in 58m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f380b566c0be470ba07a1fe78f2b87ed : FAILURE in 1h 32m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/89514258a23a41b1be43c733ab972fea : SUCCESS in 37m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/baaf60f9f1cf49758ee601e78eeee0e5 : SUCCESS in 2h 13m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ff7ce71b0c85404f811dc40b894f5523 : SUCCESS in 2h 09m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d69793d61104236afdaf40a54534d38 : SUCCESS in 1h 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a75822d263754d7da15ed8a63bab3cbd : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":83},{"id":"5b7e763d821928a455eb2bfa11f0f86a55284479","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 13:43:50.000000000","message":"Patch Set 83:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2504/.","accounts_in_message":[],"_revision_number":83},{"id":"cfdbd3859f86d6b56a7ab9fb451770fe5619a306","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 14:11:17.000000000","message":"Patch Set 83:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2390/.","accounts_in_message":[],"_revision_number":83},{"id":"42fc3aebf3d44e2837f578aa1d3f6ab60f498a6f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-11 16:34:59.000000000","message":"Patch Set 83:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/91995241dc7a433993c902b8e8c26f1d : SUCCESS in 42m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/32056d96a5854587b8db0f0aa3523bc5 : FAILURE in 1h 02m 34s","accounts_in_message":[],"_revision_number":83},{"id":"f2c5e3f5385b396636a378f65776e4a172a81074","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-11 18:52:18.000000000","message":"Patch Set 83:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 18s","accounts_in_message":[],"_revision_number":83},{"id":"456ab3ff6524d986e0f29a80a93e75ea4534d217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-12 09:35:16.000000000","message":"Uploaded patch set 84: Patch Set 83 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":84},{"id":"b0c7d556b7121efe36eb657689e4234d865ac3c5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:36:41.000000000","message":"Patch Set 84:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2526/.","accounts_in_message":[],"_revision_number":84},{"id":"3c636274b0ddcf6bef7ed95bae336e2aa9e1fccf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:36:47.000000000","message":"Patch Set 84:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2412/.","accounts_in_message":[],"_revision_number":84},{"id":"c958b3f0e87f020d2ac2312d655152c94a7f33d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 09:37:50.000000000","message":"Patch Set 84:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":84},{"id":"367ce55ddd252e50e757c8410b6f30e82bab9b83","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 10:07:51.000000000","message":"Patch Set 84:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":84},{"id":"0bd73a8290705c9c30a24579f04985588ff81f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 10:12:50.000000000","message":"Patch Set 84:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d208f463887c4e16b9c2412f207f70c2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bdefd9c4e5c146ebafe6029913dbd09e : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c1ce3e03d704f46bfa8835e0ba63861 : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4030e926c2fd4d4d98fbdd28165e0135 : SUCCESS in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"4500c65da73177a1f361d153af71bc194300ce34","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-12 10:21:05.000000000","message":"Patch Set 84:\n\n(1 comment)","accounts_in_message":[],"_revision_number":84},{"id":"1706efae05008267c72b0ff1eca75724f1c694e2","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-12 10:21:17.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_13915\u003e.","accounts_in_message":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"_revision_number":84},{"id":"6ce99a6f7895bf1c747cc5c23aff21e738a9944d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 10:27:39.000000000","message":"Patch Set 84:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":84},{"id":"5c3d2e186a0ef3d36e26947f8201ed7f61432247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 12:22:28.000000000","message":"Patch Set 84: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d9a5ba915ad400a854e4f5da3f1b841\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0135035ecad04633828210ca502df678 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca59000c87de414ea39d2dc590540b8b : SUCCESS in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd6325e417af4db5966cd2861d22e1da : SUCCESS in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58baef7c13b9448f854891da8766e2aa : SUCCESS in 17m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/dedabcc587ed4c7eacc7e286a887d75a : SUCCESS in 1h 04m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29b9ee2a118640fe861d150f573d7ef3 : SUCCESS in 57m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16c112873dec45a78789348971792d5d : FAILURE in 1h 40m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/377844912f69411c8fdcdc3054ad6a47 : SUCCESS in 40m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fba625faefdb4b32803f0f30124784ce : SUCCESS in 15m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f06500460ad419c9ba0030706f4bd23 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/702803ab96b042c6a7b2310a2958e27a : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dce9cf864bd149698bd8e63d72241384 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/23170e048f504b61a045b80080aba484 : SUCCESS in 5m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/feb1126fed92407686dd578cab366f72 : SUCCESS in 8m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51cc7c55c61a4483aec011912620e6ed : FAILURE in 46m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4c488256e74c4c13bd32c769818b996d : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58999d3911444ae68e07d79095fcdc57 : SUCCESS in 1h 46m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/addb6ee5cfe64632afe3da2ca2582dc0 : FAILURE in 2h 40m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95fa795c98de40098b4e733bd75567d9 : SUCCESS in 1h 51m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fa5b7d463540447b80eb535c3a3ae295 : SUCCESS in 1h 56m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/738397955694423797571cbfb989d23c : SUCCESS in 23m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ad07b0ef0e54afaa50df0127d91c37b : SUCCESS in 1h 18m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/155027db2aa54685bcf4fd915d5a17bf : SUCCESS in 1h 48m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4012167eadeb483cbe31c7a4e1c78375 : SUCCESS in 46m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28986b7d820a433d8d2352832fbe0534 : SUCCESS in 1h 49m 46s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/459f6672c3aa4a77b37cbdcaa8e332e3 : FAILURE in 2h 05m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3a24895f3304cbda3ff1ef0ea2542c6 : SUCCESS in 1h 59m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21751327b27040e9b21d76001cc25163 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":84},{"id":"bb3775341f6750fcb41535871cb206f97beff666","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 14:19:32.000000000","message":"Patch Set 84:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":84},{"id":"b1b151cdb515d94b8b0a95fb90ee831acac30d4e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-12 18:23:29.000000000","message":"Patch Set 84:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b413b6b39d1b4ba2850716ccc3da5aa7 : SUCCESS in 43m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ef8755c8cedc4d0198debaa531bc64ad : FAILURE in 56m 33s","accounts_in_message":[],"_revision_number":84},{"id":"76d246d4baa1895fc41ee0ae1de6a4c1d6205547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-11-13 10:22:02.000000000","message":"Uploaded patch set 85: Patch Set 84 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":85},{"id":"e223080c17cb844ff4fa84808188976332c54b66","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 10:25:46.000000000","message":"Patch Set 85:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":85},{"id":"755d1b6559bf2e5126e50655de41901688d6826a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:00:20.000000000","message":"Patch Set 85:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02cd14343c254b8cb688722191d215dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d7ec1d14c0341e5befe59edb8a8c151 : SUCCESS in 28m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65064bd27bac49d681a271dff7724bc2 : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ccd9a576ae946fe9384e1f3dc558eaf : SUCCESS in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"f18fb144b6fe8df88d93cbf1116b225e8a724f49","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 11:21:22.000000000","message":"Patch Set 85:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":85},{"id":"58fe79199117c78e278afe759fe30dddcbfb7db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-13 11:52:09.000000000","message":"Uploaded patch set 86: Patch Set 85 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":86},{"id":"4a9e2b523581b9b32a3eee5f40605bb9405341c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 11:57:16.000000000","message":"Patch Set 86:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":86},{"id":"c02ffd8f9b1e24becc24595c4bf223257fa3ecf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-11-13 11:57:28.000000000","message":"Uploaded patch set 87: Patch Set 86 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":87},{"id":"4bfa23c19c0f86ee1e6f8daf9a52c8d91bf43d50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 12:15:10.000000000","message":"Patch Set 87:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":87},{"id":"c050afd86a9601b8a3a088219be77cc865559f99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 12:22:57.000000000","message":"Patch Set 87:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c7e84f3c4e40ed81d4ece7752141ea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecec369299364d089713dfb4914f6f64 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c70b179e5a4c48138fd3968cd030074b : SUCCESS in 22m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6316d1e522d34a18a699938307f98a15 : SUCCESS in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":87},{"id":"bd6b86bd1ad62232c57c24709e69f2a9f13f3c35","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 12:51:31.000000000","message":"Patch Set 86:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":86},{"id":"0bae56b5aa12b1c8435e725d683ac4b73c306ee6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 13:27:33.000000000","message":"Patch Set 85:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2436/.","accounts_in_message":[],"_revision_number":85},{"id":"bf63a74318491ca83f3bc6eebc2823c5cd9b8fd2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 13:28:55.000000000","message":"Patch Set 85:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2550/.","accounts_in_message":[],"_revision_number":85},{"id":"a999d6c7d3c369764434c1861dca14cb3a6a4b5f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 14:49:07.000000000","message":"Patch Set 87:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":87},{"id":"768f1e1abfc6913dfbb2ae2718becb7e02cb2990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:04:41.000000000","message":"Patch Set 87: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cfa916678ec4680b5652c636eb7723c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d829fffa8f4ac79cddc9e45fb34d2b : SUCCESS in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f533f941f0041b3933688582be4abee : SUCCESS in 11m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5aa536b226c64eeea88a5c3e18fb0d73 : SUCCESS in 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851d163c9896433d87747007cb0bbd4e : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ebeca089fc5745b7b778bb6ac23fe8f9 : SUCCESS in 1h 10m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bd2fbfb9833490bb42e3b0524c8b33c : SUCCESS in 58m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4746de453fe46698ec64158ffec2bee : SUCCESS in 1h 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56d265745fb64a4bbb6475cf965a81f9 : SUCCESS in 1h 08m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69ef4d95956e4e68873b8796df645009 : SUCCESS in 17m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/94e53e68edd843ef9e00c4754ff3f83d : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7c674676c8024fcd9d5a43d0af26c259 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ed2c32cbd9846669d15774029b51127 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbf29b2783974b3982bd15d4805bbb9d : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/93ab6c7effb94372b7a9f9a4c014c5fb : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/48932df6f67a4fbe908d703ba70d117d : FAILURE in 42m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd661861ed594465a718a4669edf0e55 : SUCCESS in 4m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/860bad33811b4233a7ffa053917df6d0 : SUCCESS in 2h 09m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b142919c8532427a882765fddb921a33 : FAILURE in 3h 02m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/862e4225190a458f9a1d17c605a29466 : SUCCESS in 1h 52m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6366884b0b9f40c78201f69fb6107c9b : SUCCESS in 2h 34m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bec716708e65491d88ab983e23ecfb47 : SUCCESS in 35m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/364c2594369949e092f0962b2031b79e : SUCCESS in 1h 18m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/240d2fa21bae4ccd8c4b2ddc9ea08a98 : SUCCESS in 1h 58m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92a3b9fbc2bc42cc96d46da10df359ff : SUCCESS in 56m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/171a26e49ad24ee3ac3174076956e02e : SUCCESS in 1h 53m 23s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/637875f84cb64fb586a844920f627b1a : SUCCESS in 2h 11m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5e442ad2db948d184985b02ffa2c161 : SUCCESS in 1h 59m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86881e2f256a4ec5b4d28c89a4ee54dd : SUCCESS in 1h 49m 31s","accounts_in_message":[],"_revision_number":87},{"id":"2757400d50c30a8bbd7a9bf3eb73072b7da285bf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 15:13:15.000000000","message":"Patch Set 87:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4766d4e1865343b0ab45356abb90bd1e : SUCCESS in 54m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8317badb6ecf4869abb4fb981706f94a : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":87},{"id":"ee83c98ee5e04cbd24fbc6777b1993a8805ddbc5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 15:19:26.000000000","message":"Patch Set 87:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":87},{"id":"9aeb1fc9717c4aa6ad81f1568e3fad633a22a357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-11-20 05:54:45.000000000","message":"Uploaded patch set 88: Patch Set 87 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":88},{"id":"eb70354e0b94694eb97cf10be101887a0218c4b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 05:55:45.000000000","message":"Patch Set 88:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2525/.","accounts_in_message":[],"_revision_number":88},{"id":"fd260a725f0adf3d14575e260eb626c7cc8ddaf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 06:27:38.000000000","message":"Patch Set 88:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac151956f094776b96751d8b6b5589a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1c09023a5964f70b7ffdab01f685788 : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a52068fcb6246f4a75b09329ad15657 : SUCCESS in 29m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24f16784225d4285b8183b3b0584cabd : SUCCESS in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"bf773cd4eb4c2c8874165cc4451423fe86aea5d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 06:53:25.000000000","message":"Patch Set 88:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/24/ : SUCCESS","accounts_in_message":[],"_revision_number":88},{"id":"78e3a7516b9952dfa704a189ed6c58b51a595989","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 06:59:57.000000000","message":"Patch Set 88:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":88},{"id":"9f0c0ceb5c3e40e8b58c377f1c0a3083ad243777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 08:25:06.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c010c1aebfee4d17a8ea218ecfa92131\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2af5424fc148729e32cf271b796919 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb4a03d329e24f1d9d73d7af3c8bfb50 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcefe5ea21df452988f1512153549991 : SUCCESS in 15m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0b8b55981d42d78c3218b8b5a9159b : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/852127e450224607801f8d28be41a772 : SUCCESS in 58m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/087d49b1c30a431ca55bfd3d6397e66f : SUCCESS in 47m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/329d64fa30d64da3aeda6a6d2e610998 : SUCCESS in 1h 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f43282c606b4100b38ce3736f1b3afb : SUCCESS in 49m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77fdd4c13fec4de3b328089ac80faeb5 : SUCCESS in 15m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ee1513cc6d849d197c3df3b94cffbbb : SUCCESS in 20m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e21a9cb51d124322a314ea8613fb1380 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c08d9da026a64b7f8c21b72ccf728cbb : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/384e7cdd731b4430bd927631e1bca196 : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e3aa64ddc9b4d3d9b7b5fbb2a8aca7f : SUCCESS in 8m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b64a73c52c88497ab7b6cd1a42d643f3 : FAILURE in 27m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62e02da121a941d6a1f8de3543a957f6 : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/183392d764a44dd1b084979d35b194db : SUCCESS in 1h 27m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5416358c58344eadbe442470d811fdeb : FAILURE in 2h 24m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5f2cc09de6ca42df8d984e2c3033d664 : SUCCESS in 2h 16m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fcd4cfcf5dd54d859a0f1bd8ebd5579c : SUCCESS in 2h 02m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9d9abe393ee84ae59398b40b22660c28 : SUCCESS in 38m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40b60fd9ccac46d5b844f937115e5754 : SUCCESS in 1h 13m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bab6e8cb2ad24a39b4d747edd9112e73 : SUCCESS in 1h 46m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/27247fbaa65b4659ac1ccef1257201e9 : SUCCESS in 32m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4bd571deaa6467b9d448cf9a9420e30 : SUCCESS in 1h 26m 49s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9817dfe966284ba8b0dde23344b1931b : SUCCESS in 2h 16m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8a0c784b1904bcc8c028562d7e55356 : SUCCESS in 1h 42m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3a6625bede8430c8b2cf2bf41aa75d2 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":88},{"id":"41c4a9a96055333d24cc4b9a4dce46f26e399c9c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 09:00:02.000000000","message":"Patch Set 88:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1bea5d0bc08344328811d39564cde5e9 : SUCCESS in 48m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5fc599618a8e496c84e1f99dd2b88a1f : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":88},{"id":"04352de97d4d763ff0c66cd88e23cee8f7d202d6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 09:05:27.000000000","message":"Patch Set 88:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 49s","accounts_in_message":[],"_revision_number":88},{"id":"580b7788f4434046a882d625eee9d533a18e93c3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 14:58:44.000000000","message":"Patch Set 88:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2669/.","accounts_in_message":[],"_revision_number":88},{"id":"3849f0b5005db3d793e99843e322a012e5cea97a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-21 14:14:54.000000000","message":"Patch Set 88: Review-Priority+1","accounts_in_message":[],"_revision_number":88},{"id":"aff762c9f80ab49ed97a1e5fc81abf002019d2a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-12-04 11:29:34.000000000","message":"Uploaded patch set 89: Patch Set 88 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":89},{"id":"b8c94a23ac4ea9765fc587abfb0bd6141483b1fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-04 11:30:21.000000000","message":"Patch Set 89:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2774/.","accounts_in_message":[],"_revision_number":89},{"id":"dd49552807435df155b88b9daf330fe36c186196","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-04 11:30:21.000000000","message":"Patch Set 89:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2626/.","accounts_in_message":[],"_revision_number":89},{"id":"081ce10f68bbd822d69ff0908436cee5c3c277a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-04 12:25:50.000000000","message":"Patch Set 89:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/123/ : SUCCESS","accounts_in_message":[],"_revision_number":89},{"id":"81cc444f0830164a8f1105c482f34b89db076403","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-04 12:30:08.000000000","message":"Patch Set 89:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":89},{"id":"5958e358c279d1f20e68bd29a1c22b97ddfdea4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 13:37:30.000000000","message":"Patch Set 89:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2ca1ecb04fe4ad598a73c5c3ac40754\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d37e85ac2274d38a1c5fda1d07bd350 : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fac665f4ffa64f91874c03bfd2265c97 : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8e99155da85f49e8a99e69a7f2df9d9a : SUCCESS in 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"225e1d17a092a3bd3284584804a4c0091410e199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:09:59.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ccbbf690974bdf864c166049ebc49c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/213eec4a3ed44ab1a0ea27d72a5aae5f : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b7b8aea75cc4f3bb4bd43e1613a83bc : SUCCESS in 7m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e497086ebba416a80aa1b11b402b6d9 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70990aa56a9423799c3fe72638db8e4 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/7bd6af1d7b994d94a39b7bbab4bc96ec : SUCCESS in 58m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0f49e5a0a574ad4a16fa7617b0fc811 : SUCCESS in 32m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b99fa295e4f4d6aa937cf88622e2a63 : SUCCESS in 1h 54m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e66a43ce7724784bee56c9fc3d4dfb3 : SUCCESS in 57m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddbe9ec268fb493fb2b5be0436f8385e : SUCCESS in 15m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b2083af386d4868af15c946c3eb1725 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2485fdc623f441bfa0cc3325128e26ff : SUCCESS in 3m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1bdbb1bc79e34ea0ab15e753448fe916 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d906342197884dcd9041990f36eea173 : SUCCESS in 4m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a15530013b54453a8fbf52f1d42f0e25 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1090107df2ad42b19ad89fa45be76613 : FAILURE in 22m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f2dc5c09dac408c829b23517bab0216 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/162c09fec2a64bdf9a4b06aed3fd2a8d : SUCCESS in 1h 36m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d25fbaaf068434e8ab8cf731ce5b389 : FAILURE in 2h 21m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3cfc9466cc92447ab8b10888222b069e : SUCCESS in 2h 00m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/decbb96c4f5b4c3c964de930ca8dc420 : SUCCESS in 1h 48m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a889015133ba46ec97bcb1da500023d8 : SUCCESS in 19m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cfcf0f8d6d44119a3bebdead957294b : SUCCESS in 49m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85755d73a66140ca9e9bdeff5436f85d : SUCCESS in 1h 29m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d35610847ff64aa5a6cb2fd0d3bd49f4 : SUCCESS in 24m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1d224cebf3b474597b618e18fb991b9 : SUCCESS in 2h 01m 39s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/17a03012217d42e3a71cd36af6992e26 : SUCCESS in 2h 27m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ede67c6cad84798bd2017c7c37fedef : SUCCESS in 1h 40m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93b3acef4cb145afb8702ba20a972613 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":89},{"id":"05cbb3e4a5bf15d79ad244a78de91e3bbe3bec15","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-04 14:17:03.000000000","message":"Patch Set 89:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_12_2025_14_10_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":89},{"id":"a1c20614705abb4759561c461d11a3b0fdcada33","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-04 14:22:44.000000000","message":"Patch Set 89:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7c16cdc5ca8d417cbde8e52d4dc9b4b2 : FAILURE in 1h 05m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9ecde432506b4648893954680481599e : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":89},{"id":"8bf7b74ff586bdf6c1f1c8ec112328325f8effb9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-04 14:26:18.000000000","message":"Patch Set 89:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":89},{"id":"ccaa88a904ed3e40a4030f1ace299499cc0811cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-12-08 09:59:23.000000000","message":"Uploaded patch set 90: New patch set was added with same tree, parent tree, and commit message as Patch Set 89.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":90},{"id":"50cd2f20acbd1b367eb0eaead43ebea1fec1ae85","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-08 10:00:15.000000000","message":"Patch Set 90:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2794/.","accounts_in_message":[],"_revision_number":90},{"id":"9c0297acf1d91c9c4f6ad6129db21afc33012aa8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-08 10:00:23.000000000","message":"Patch Set 90:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2647/.","accounts_in_message":[],"_revision_number":90},{"id":"52d3ae6a1cb5eb4bb5db938fe007cf79ff2e50f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 10:33:15.000000000","message":"Patch Set 90:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bf06d1bc494932b68e79a53fefb068\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b5b0a0b67154f6586c52683f5d325a7 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fea762bb2de94bada26e67922bf7782d : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0a92bab208ca4aa0a1e59c1d182f7b6f : SUCCESS in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":90},{"id":"28cfeb6310e3408eb98088d099b24ce6f848988c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-08 10:51:29.000000000","message":"Patch Set 90:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/146/ : FAILURE","accounts_in_message":[],"_revision_number":90},{"id":"8a48f97c64742e47a5a28798f7ed0aac08936d33","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-08 11:03:34.000000000","message":"Patch Set 90:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":90},{"id":"28eb530e7e70f1a11e6d008630bec7c3fc8e3d9d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-08 12:56:30.000000000","message":"Patch Set 90:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 34m 48s","accounts_in_message":[],"_revision_number":90},{"id":"ae9b546bfac446d15d36c031ecf2d1d4914e10b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 13:12:49.000000000","message":"Patch Set 90: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c60ebff55f4b0b99457db977b88d81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63016de9d894959beac98c1866a5752 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/780c533a75034250b5e7af1aa5a465ea : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d113e921e124c679d86b163888b96e3 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a21d7f979c4e99892180b93be343fe : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/7d08d56556f246aa9b1bf1fb4b503b47 : SUCCESS in 34m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ae0ff03a0064d8ba37157499ada1313 : SUCCESS in 59m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d88e33ad9cb433f8af2f88f6fe80779 : SUCCESS in 1h 28m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba8f5a6286534585972959a2166edeba : SUCCESS in 1h 02m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b01fbc6ebc44e8a2b4e25708f427b8 : SUCCESS in 11m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/519c50f162cc4bfc9b342633b8f23638 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f12489b85c204725958955f42c838567 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c33c6c8bbc5d45258fb6d67ae18922c2 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81e506de1afe4b84bddde1d0e934d7e0 : SUCCESS in 5m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0d052c5d78cd440aa0cd271de132a056 : SUCCESS in 8m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93f031a7fac443b993647cb0f7034af3 : FAILURE in 46m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5923cb499a464b02a36870dea8465a1a : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f0800bcc6048456fb4f571a9df771b0a : SUCCESS in 59m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41a9e363fd9e49258d1758dd38458080 : FAILURE in 2h 51m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f43557a42ab84d85b8afc6501a0841cb : SUCCESS in 1h 52m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/88addc469939492cae1abd6e28f7fb41 : SUCCESS in 1h 56m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/90614039a4c44207903216eab6b4a22f : SUCCESS in 35m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52afb68b29bb4bc79c21b8720c79aa9b : SUCCESS in 1h 03m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0df41f4cfe134c229ffba0dbc748b817 : SUCCESS in 1h 33m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7c443d618bfe442085459b1ce6b14f0c : SUCCESS in 23m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7f1ded5a21f941f98776057c4a5b6fcf : SUCCESS in 1h 51m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b016d54e17184767ac2c21e8aec7c5ab : SUCCESS in 2h 04m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e9eb08d6e354b35b5b4c1e9f59fdf6c : SUCCESS in 1h 59m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53e8f90d04f74f55ac485ef63bd217fb : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":90},{"id":"d5b63111696d193efe313eb9ad3a62e362749ef6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-08 13:19:49.000000000","message":"Patch Set 90:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_12_2025_13_12_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":90},{"id":"375d25c25453945b4ae7ec72f0b169f364f61838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-05 08:30:10.000000000","message":"Uploaded patch set 91: Patch Set 90 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":91},{"id":"6e176481f7d1518ce255132bd9e6f773f7cf8d0b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-05 08:44:32.000000000","message":"Patch Set 91:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b804e79e027c426a90b52f285dc11bae : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d1aa3dfe04084e21a582e9316a520a7c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":91},{"id":"4e0b462c1317d833153fd422e57f84394a38d2eb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-05 08:53:15.000000000","message":"Patch Set 91:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2c2adf94d2e0486f97ce2c989797d119 : NODE_FAILURE Node request 200-0000942308 failed in 0s","accounts_in_message":[],"_revision_number":91},{"id":"f64558161f2fcb165e51ec94dbc5e004c2ede01c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 09:08:49.000000000","message":"Patch Set 91:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23134fa3f79d48fab182ef3817429b8c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aca2d8ccf4474efc9e00d3e49f51ac08 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e52046d0c9d44db8289285a6db8e1f8 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18fc2467bc2c41c381969e5c3f02ce67 : SUCCESS in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":91},{"id":"72e369e1a028b168ee1f3598a79cc74e83a874d5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-05 09:32:04.000000000","message":"Patch Set 91:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":91},{"id":"90d82895870559ac51aa878709c438564f279f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:47:34.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc41ead111234d33bdb6014ffaa0b060\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e397b093ec44ac8e0edf4db785465e : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e209a6cbd354d7f870a770d37e61f0b : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf9cc704c2a54a838edf8155d1482114 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a2bc455f204af0afec7a32d1589a0a : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/954798e11f7a41ceba5717d810c6db4b : SUCCESS in 30m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/675b7364f0a74c4c98d13100c27b1215 : SUCCESS in 59m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/539c9044f94b49e59b17dcdb339f678b : SUCCESS in 1h 56m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/411bb39d96cd4b7ba706e3ec10830a85 : SUCCESS in 33m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/905c73e7383247b99fe8a0556129bdd3 : SUCCESS in 16m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fad53ae8043340cf9a26f35b90ed2fc0 : SUCCESS in 17m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f8bfe30191743a3a0ceb4a4582e7800 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e867ef7c5c24113b4a5d8c03e667cbb : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35e0da0f4c1547028bb30f592b576b85 : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a48034d069104931b555634562642ff6 : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b26017b583a54a08866fe0f875ac9329 : FAILURE in 38m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/536054c4e1004307ab6918f856075d3b : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/529f3421c08842ac813ee880b810a8c6 : SUCCESS in 2h 07m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/651d8bf9147f431da43355cffd626bc6 : FAILURE in 2h 47m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8b9b2acbfff430993df27100ecb0f0c : SUCCESS in 1h 46m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/eb3fa62e10c2448fa4a865a1ca0d43a3 : SUCCESS in 1h 37m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2049ee803b254541b48d3fbcfc5f0f94 : SUCCESS in 23m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8d2fd6ed25d1464a8ed212c0457cdd14 : SUCCESS in 1h 11m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1df85186468f42bdbe2d1384ddf38208 : SUCCESS in 1h 11m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0663ae3986e04e7b9b6573c1a14e350d : SUCCESS in 30m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f52f5c2bfb9549af9e0250741c5cdc01 : SUCCESS in 2h 14m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f7101cd40b0e4183863b6c32cb760ae0 : POST_FAILURE in 1h 54m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f64a1a5562ce4e2f9c26082db354c017 : SUCCESS in 1h 37m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b06eb5bbda84342a52c455373e4ac6c : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":91},{"id":"1b02956036a57c31c8d3204ac448199c6a078a77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-05 12:07:00.000000000","message":"Patch Set 91:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_01_2026_11_59_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":91},{"id":"960b04fe6c8b0c98e776c96067710f4bb40ff2e0","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-05 14:00:27.000000000","message":"Patch Set 91:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_60/index.html : SUCCESS in 329m","accounts_in_message":[],"_revision_number":91},{"id":"0dc9b520529881efc74b056666e61618b1d9c4c6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-05 20:24:18.000000000","message":"Patch Set 91:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":91},{"id":"5d49f1b0f1daa71098761ddc96d064d30788bb65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-01-06 10:23:31.000000000","message":"Uploaded patch set 92: Patch Set 91 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":92},{"id":"d1a083506fe44b8cd2a0a4f4159ad5a80da95d51","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-06 10:44:36.000000000","message":"Patch Set 92:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37acf24d603c4452be283ff26e430cdd : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4710302e3e57402dad4957f1ee7f77a7 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":92},{"id":"cf056ee9348fd6b496cfcf91e051579dfff446db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-06 10:47:31.000000000","message":"Patch Set 92:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/59cf6479b5f4433c9b14589af220da08 : NODE_FAILURE Node request 200-0000942331 failed in 0s","accounts_in_message":[],"_revision_number":92},{"id":"7eb57b5ed1c239da271ddbfe516466c2ee71cb93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 10:51:33.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05663d37102d45d8a915d48ca8d35aa6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54c20c5f1a5349c8a1546b87a4e004b7 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/125e561b5b4a48bc9085232c589bb989 : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/528b7d3e044e4e9b817c72b3ee73a9ce : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"a8c2be206091a25d229bcc32754c838480e7aa1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-06 11:26:30.000000000","message":"Patch Set 92:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":92},{"id":"051ec6ff794f27dad4bcf49df51abe44a1a85504","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-06 12:14:01.000000000","message":"Patch Set 92:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_85/index.html : FAILURE in 110m","accounts_in_message":[],"_revision_number":92},{"id":"9679d80d2a2ed7465106d6b9a247c7923f94a6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 13:36:55.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512eaf544e0d446cb0d82fe17026d5d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2c84712d924c81bfae389abc706ebd : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecc24a5fa11f485f815ce0a523461e32 : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0bd655aaef54ceb86c09467e9bdddbb : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfc87ea96454e1f8668caf5d3e1134f : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/aff340f009c64880b603e67ba0d71d18 : SUCCESS in 58m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b208eca409f34aa892d7e8ca90c41cd6 : SUCCESS in 59m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79b1cc269f71482cbef40f7ae84adae1 : SUCCESS in 1h 19m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5557ea9413214ed48ec8691b539c229d : SUCCESS in 1h 04m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01f6d65cfb9041cd845b4853dc76bcbe : SUCCESS in 15m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86acd43f818d48a0b972d8b87bd88769 : SUCCESS in 16m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/56e1f9c372844d6fa07829a0a2d0ea21 : SUCCESS in 2m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41fdd9c40425416aa9a04baf9109ce59 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bda74010fec34dc1aab0cad9449fe300 : SUCCESS in 5m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b2a0cecb3ea46bca4ae1a43ea90589c : SUCCESS in 10m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3996b5ea5294446e81384cb0a4281926 : FAILURE in 38m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbd7e0ecb9d840a2bdb77001c67906f4 : SUCCESS in 3m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40fa201f9423400ab312f0e53b975ea5 : SUCCESS in 2h 02m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ce4afba04c647d5a0741d2965093287 : FAILURE in 1h 39m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0bb97f46c9f74d558d00be82d7289a26 : SUCCESS in 1h 55m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1413c4245f3644d9b5f533714956f241 : SUCCESS in 1h 41m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c0699d0cce34d3aacf88b6c342a6977 : SUCCESS in 41m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5cd456c85bc44e77b3a84a4a08c071cf : SUCCESS in 1h 11m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9082e49c391d40d3a16ad9de73482055 : SUCCESS in 1h 28m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c61203d3cb334a0981181b3e30b289ca : SUCCESS in 54m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/109acc62c58e433fa418ddb659f36533 : SUCCESS in 1h 31m 21s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/aba77455192742f896898badd92c04ce : TIMED_OUT in 3h 10m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16846f9b9fbd458abc89724b11d5ec82 : SUCCESS in 1h 01m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3da14a9cc141490c9e13701c7f9bd0e6 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":92},{"id":"ac735fd8f65aa52d246302ca7b49ddeb4ddb01c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-06 13:43:34.000000000","message":"Patch Set 92:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_01_2026_13_37_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":92},{"id":"2efdbd4f732d88963bf7a50e64fa421ed494c34f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-06 15:36:04.000000000","message":"Patch Set 92:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":92},{"id":"c2f23677003f46d8d45caaf5d52b8a52c19f88ad","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-06 22:06:58.000000000","message":"Patch Set 92:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-92-fc : SUCCESS in 1h 3m 44s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-92-iscsi : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":92},{"id":"0244855627ca069c59d97f1e3f4f1c7fe2e17c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-14 18:06:26.000000000","message":"Uploaded patch set 93: Patch Set 92 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":93},{"id":"ce1401fc526c3c0bb22d3bc1303aa6998c2a9cde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 18:35:27.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6221c154974667ba0b32750ead7c68\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/996ffcd67f9d406f91250d186bcf629c : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/437cf83e93b64fcba7db51cec6411144 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b421c7695d174b1ca1f9c9f613498a3f : SUCCESS in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"a978a47f767bf4cc60ec2385ececda0f0deccb2e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-14 19:13:42.000000000","message":"Patch Set 93:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":93},{"id":"b5532838c6aa03316feb949c7bc4abe98de5093a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-14 19:50:13.000000000","message":"Patch Set 93:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c4cae1437d09494c87a2e988a657606b : FAILURE in 25m 25s","accounts_in_message":[],"_revision_number":93},{"id":"eec8c941eb4a305fc7c5d20877bb7845e0c09686","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-14 20:16:39.000000000","message":"Patch Set 93:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/190214e366554d2e92b4896a334064c6 : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":93},{"id":"db087da98e417fee481b28b28f3e5d2f73fdbcd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:58:59.000000000","message":"Patch Set 93: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d19be7ed508468f819d9f151ca6590e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e660dbe1f24f959ccc5e2b2fa219b7 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34f5bc0ecef54859938f64b3c97edda8 : SUCCESS in 8m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89cf300827634a5481458a0770bd8f0d : SUCCESS in 15m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dad762285fb47248bf2381924af4791 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/68208cf8fd83441e90989642c8ca4f5e : SUCCESS in 58m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9f040cbedea42cca02f9fe30a3e7f3f : SUCCESS in 51m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fa719718fef4c02bc4c1fbc39e95f1b : SUCCESS in 1h 42m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2df981abfb874c4f90dfea457c95c83b : SUCCESS in 1h 01m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ddf40afbff44a38b567d56d4a9d2dfa : SUCCESS in 15m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de6ae508642043c7a8e7a366d47bc5db : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75af6cf90f4d47edae3b51862e8b1102 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a6c928421884343b92c9176340a6a09 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e38ea4ee75a4464099a8cab9bbeac070 : SUCCESS in 6m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/96a79cfbd9b0461a97d31665c382c671 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b73795587ad4553b9d7bdfba287a086 : FAILURE in 31m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca8bed72b9044335917cf655b7a1723b : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/143ba042067b489d9bb33c60da269556 : SUCCESS in 1h 45m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/89bdd7759b024f769953c2bd1fa07338 : FAILURE in 2h 46m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df934280f30b456c832f2babcb855625 : SUCCESS in 1h 42m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c82bd977f3a74f87adfd86313470493b : SUCCESS in 1h 46m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e14eba3a5bcf48b3b160af1c8c6086f0 : SUCCESS in 36m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e59a2f06068f4b85b7cc6597ebc6798a : SUCCESS in 1h 15m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/49ac20b2683d4472b8955eb3a335c2a7 : SUCCESS in 1h 27m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81d59ecf847c421088734c8a5d95e462 : SUCCESS in 39m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e2d5f76e8c384dc49ff7c1ca4ce5d3a0 : SUCCESS in 1h 47m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/27e093f88d574c039d08cea5efbd5a56 : SUCCESS in 1h 51m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a874988fa7b409fa1922727ca3632b1 : FAILURE in 1h 34m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a1c825f93934a50a8c2c229c25cee95 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":93},{"id":"5927977f2dc52521f12ce6e96c597381bbaff749","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-14 22:15:33.000000000","message":"Patch Set 93:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/217/ : ABORTED","accounts_in_message":[],"_revision_number":93},{"id":"a91dbbc644db60bac3a6cf4811ed4378b64505b4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-15 01:06:10.000000000","message":"Patch Set 93:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 34m 52s","accounts_in_message":[],"_revision_number":93},{"id":"1ec58395fb982d0d69cb58f36b5220c18fad988c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-15 21:15:32.000000000","message":"Patch Set 93:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6e4e126ca5f74154a5bf5b6b7c0cf9bf : FAILURE in 1h 11m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/609ab2b98b9a4b92908512ed8aeb686c : RETRY_LIMIT in 15m 25s","accounts_in_message":[],"_revision_number":93},{"id":"5e6440e055fcfb49e532fe79e78b33e26d4e17d3","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-16 15:28:20.000000000","message":"Patch Set 93: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":93},{"id":"d99f436d9167dc2d0f3dbaa75c3637598cfd9a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:25:21.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc67fe00c5e548428145fce40c934881\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5533b3eab5249b99f751a1bf96a854b : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f310929cb43417a97e4f79e402a600c : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e490c98aba8c4f14b525b464cc1cc5c2 : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"4baf4d114819cdfd238edf401caa63d23cd9c646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:13:54.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e593c4bbae754b6e9c807428c95c7b1d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d03cacc51c8c4ab9b985548c685ec454 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f12bc3b01752411fb0c07d62806ea1a0 : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40023114e84742c183a1c5e1ae834309 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0df4740a5e047648c0eacf02f5c724c : SUCCESS in 15m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/de7bb2f6c8464b73ad4af0c70b98c538 : SUCCESS in 58m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3b96550536049cd89727f1d451ecf47 : SUCCESS in 32m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d194cb29b564c14b7040fc704aefc83 : SUCCESS in 1h 22m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6a9f784c46f4ea095d00a97c6c42cfd : SUCCESS in 1h 02m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4fc2fb6bdce4785b9fc2fa6be95824c : SUCCESS in 15m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f9ba26732ede4ad8a40222c604f318e0 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ee3123dec454e18901e557cf1e42376 : SUCCESS in 3m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05152d413a3047968006b8ab05a22a45 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f86dab42bd544c8a8d427ac92d6d988d : SUCCESS in 4m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bca41adcc6d54478928cecf88b63a2ea : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4b99140a209143719b31faad4e10acbf : FAILURE in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/28394b8666934bc5a442a0daf371c77f : SUCCESS in 3m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab3227f8d7a84638b2f5528edb609559 : SUCCESS in 1h 56m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/97486baab42741839126185e9cc8b2af : FAILURE in 2h 38m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e671a9c912684d8eb14fcfa00bf8f16d : SUCCESS in 2h 16m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2a66e2e6a742499295609e111c7d6049 : SUCCESS in 1h 32m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf2ed2dc89a742068368ea563b72bb57 : SUCCESS in 19m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efafbcea4a5742b494669d6a14bd4b14 : SUCCESS in 1h 12m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a8f945294384609ab378608d10c69e5 : SUCCESS in 1h 42m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f82741880fda43bab2dc5c6da61acf3a : SUCCESS in 42m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8efc61aa8733434ea8f25560bc782bdd : SUCCESS in 1h 13m 51s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4a9ca980274f4a419d9932bb8350923a : SUCCESS in 1h 55m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5dab9c5e0204955aba6afa321362a5b : SUCCESS in 1h 49m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8fdd525c15146a58f852dd82f2ad3f9 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":93},{"id":"95eee415ea2942bee8d3559d3c0b8b2e59f569d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-16 18:20:50.000000000","message":"Patch Set 93:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_01_2026_18_14_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":93},{"id":"9a03fed5dc7b62c83d849d118d555f99438ad79e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-17 03:20:29.000000000","message":"Patch Set 93:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a68f3c2f9e0943dfb56cd563b278fa5c : FAILURE in 1h 07m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/df6c6021dadd44ed9a6b076692aa7e5b : RETRY_LIMIT in 15m 08s","accounts_in_message":[],"_revision_number":93},{"id":"b66c3588b8141e87d5868b55d4532d9688feb8aa","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-18 01:36:27.000000000","message":"Patch Set 93:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-93-fc : SUCCESS in 1h 33m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-71-938771-93-iscsi : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":93},{"id":"85c7203ec2aac0a8d0c284ca340a4751f60fed1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-20 05:52:50.000000000","message":"Uploaded patch set 94: New patch set was added with same tree, parent tree, and commit message as Patch Set 93.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":94},{"id":"a144576553bf51338f84ab095ba40b30610ae6e6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-20 06:08:21.000000000","message":"Patch Set 94:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":94},{"id":"54b69bf225496d65a5b1b366a30066bd2e8842dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 06:17:35.000000000","message":"Patch Set 94:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b4fcfa69b5143989ae7a299d73b484f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8972394134134898b8001c198f72792c : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5123f130d78a441281c7d5a399d09344 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f2e1df6dac441ea90540dadd4d6c02f : RETRY_LIMIT in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":94},{"id":"a4646dafed6aa5476f46c78f77222ee780db9f21","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-20 07:41:30.000000000","message":"Patch Set 94:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/93fb50cca27d4b97a698a0415bc810aa : FAILURE in 27m 01s","accounts_in_message":[],"_revision_number":94},{"id":"ae338cafc8596748542c181b24b1cbbc6f6ab39f","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-20 08:08:14.000000000","message":"Patch Set 94:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/576e916d280b4f2b922569f6818fb2ba : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":94},{"id":"1e574e8dd28113a3515f2bfd3b404a535614c49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:26:33.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c65fb0d70d36464bb9a5b264b973f7e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b32697ad174f0da0fb7fa701a3c2a4 : SUCCESS in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d2af84b3d47440f935f3b4d6b5d0d32 : SUCCESS in 11m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09671ae55e43464db1745b5e39b8ce05 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a091b97f7965431b97e6c5fc966efd6e : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ef4441a61c5b4370b176e40369a272f4 : SUCCESS in 58m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2e52f976a054f2d8a2eb32e55f70b78 : SUCCESS in 35m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a5557509e02418d9568a0eb9b3c0d26 : SUCCESS in 1h 21m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/291404ea773e443ea7ec33b910b5cfa0 : SUCCESS in 1h 04m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a4af5dcb2c04e39a7189db0a0a23049 : SUCCESS in 14m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5decd04a15b467aa699c90a3b15b1c0 : SUCCESS in 16m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8bdd1a0f656d4f698867265de292b373 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e245f366f79d485a97fff386e92e3075 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1180f608e6d24c5e85995d10f24e7fdf : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8b7981d434ff4a7b93dcfa2e6ce294e3 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/55bab9093b434a11b83cfe8419dd6f78 : FAILURE in 54m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/754501497e99462c8e6291a5d078a359 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/96a05750ce6b4c288f9e8194b4b7d5c5 : SUCCESS in 2h 06m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6388eaac17c14c68981f57cf7b5d966b : FAILURE in 2h 26m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecaea17115b841eaa0c8233dbdd63d8e : SUCCESS in 1h 47m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/22d58d1ec65544f982107f56a34b8b57 : SUCCESS in 2h 17m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8c48866a33c1459c836b5af00f179f8d : SUCCESS in 23m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71d7abdbb8724fc7ba6df2a885de3fc0 : SUCCESS in 1h 11m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/802c56a25bc8422e924310cb3f351f31 : SUCCESS in 1h 37m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c402317450f4c1780182690a755b6e7 : SUCCESS in 45m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/289dbc98132a4e52a49b93bb5c11eb77 : SUCCESS in 2h 04m 13s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ed569394e9194448a87ea9d644bfa2b5 : SUCCESS in 2h 20m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34cdf28768e74ce1ac3ca26597110426 : SUCCESS in 2h 12m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cb099f5c85446ac8bbfd942ad787d1e : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":94},{"id":"20ad0be19d267e439a9f21762795367efc3b6537","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-20 08:40:28.000000000","message":"Patch Set 94:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_01_2026_08_32_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":94},{"id":"d2b6db8b1991fe73a62ada34b8eca534ff5da25f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-20 10:09:50.000000000","message":"Patch Set 94:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 49s","accounts_in_message":[],"_revision_number":94},{"id":"d660dc156c21bcf397537cf0277e03e581592a24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-20 12:23:45.000000000","message":"Patch Set 94:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ef4be7dd960a487e8223ca6c19b3639a : FAILURE in 1h 22m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4108e3c8eb2f49baabfeee01cc730f8e : RETRY_LIMIT in 15m 02s","accounts_in_message":[],"_revision_number":94},{"id":"c92f8f2c97cc8e474cf9754b7cc06371d2b4a6fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-01-28 06:59:48.000000000","message":"Uploaded patch set 95: Patch Set 94 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":95},{"id":"d38f5e0c6f2b3e8046cc9a3ab1be887c15389ec4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:25.000000000","message":"Patch Set 95:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/683/","accounts_in_message":[],"_revision_number":95},{"id":"caa54036bf23d9c0fcd397b13f5e8bb18134d07b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:30.000000000","message":"Patch Set 95:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/95/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/95/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":95},{"id":"d770df8607c10eec5c6e3b290213af210474c756","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 08:06:04.000000000","message":"Patch Set 95:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":95},{"id":"ee7e9c8a57549fbe07b700c393423f3b37b4c5a7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-28 09:14:52.000000000","message":"Patch Set 95:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 41m 55s","accounts_in_message":[],"_revision_number":95},{"id":"067c7a29865714b2f0570cc6556f775a82e5a1e1","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 09:25:34.000000000","message":"Patch Set 95:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/93fdaf2e396b4969ae1845a393663d4a : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":95},{"id":"79d0442d617eee6e1258b2a9c918c3bd1f81d82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 09:25:34.000000000","message":"Patch Set 95: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2700bb5a5148c1aaac4391c2992a58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5159c7917c469a83f72d0e98d08fc4 : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/466564ea8d5f464e8e79162c915e8736 : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0ade8ff721449978a46896070abb78c : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba9a7b7830b4f6bb6c343e28a23fbb5 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/7b1e4db1f7bc4c44934b5d11bad0192a : SUCCESS in 1h 00m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f466748f0f284a8399e73849199a879a : SUCCESS in 57m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90b6226bd1be46ef82209ccb9088eca1 : SUCCESS in 1h 35m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a249633b22b943d6a2d5d1cf1fa37226 : SUCCESS in 1h 01m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfa2cc05f2574498a9d88d77e5b3652f : SUCCESS in 9m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e580b89ee294880a71bbf7dacf79eb1 : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a1626bad8be49c9894d30f0ad18d894 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a013b785f75349598cccfe8201c64acf : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/393a2d8514264008af6f692fed36a2ad : SUCCESS in 6m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f9516e79df3e4a62b3d4ce2d8aad968d : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e2737f15e0c4019ba4c8dec34420c49 : FAILURE in 45m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/deb7f3fd81844a4483b468953816dca1 : SUCCESS in 3m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87b3872cad56464aba80ef6325fc5656 : SUCCESS in 2h 04m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b839188f5114dcb81e9ed4824b1f638 : FAILURE in 21m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db29b1dfe3a44bfe9850ceadc97adbca : SUCCESS in 2h 00m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2556a5edd4b84dbf8e87807ec8cd639b : SUCCESS in 1h 53m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ac372d66ad194d8c8e94b49d596ed830 : SUCCESS in 23m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a7d7ce32393462e97d56775ae618dd4 : SUCCESS in 1h 07m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b3bc5bdee0540fb8c82f97262bd3f10 : SUCCESS in 1h 40m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/299f8b628eef4a83962ef272890a79ff : SUCCESS in 42m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8bcc1c6ecd1143f5aa4c38c9cb2c6142 : SUCCESS in 2h 04m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/27df7371e696400d824e8e8215cac77a : SUCCESS in 2h 02m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d6a6f8350d64e44921023c06bfc41da : SUCCESS in 2h 01m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be7b1e2631ce450493c7a8a1388d509e : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":95},{"id":"5089167fc452cc8408c7d02bca90fd1cb30c5929","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-28 09:32:26.000000000","message":"Patch Set 95:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_01_2026_09_25_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":95},{"id":"307f674ef15d8b49188d56543dbe220b98757f9d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-28 10:11:05.000000000","message":"Patch Set 95:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c3ec43996e614b6296e84ece73dd72ec : FAILURE in 51m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e33cba5505d848ee9776704bd0c057e7 : RETRY_LIMIT in 15m 14s","accounts_in_message":[],"_revision_number":95},{"id":"3273ab22bfeda3ce8bb314e186a4bc9df73e9407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-28 10:38:13.000000000","message":"Uploaded patch set 96: Patch Set 95 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":96},{"id":"1334d1b1f15b3ed7629a66be9e2bb2db5d38a2e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 10:38:45.000000000","message":"Patch Set 96:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/706/","accounts_in_message":[],"_revision_number":96},{"id":"12cd52ea9a5a1fb7c73f8b7afbcfcd0438201659","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 10:38:47.000000000","message":"Patch Set 96:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/96/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/96/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":96},{"id":"44d67c9734ecf89941e49a47a466525f67dcc86f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 11:38:44.000000000","message":"Patch Set 96:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":96},{"id":"b17e97ff19ab9552c09b903022dd51b84db39743","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 12:54:35.000000000","message":"Patch Set 96:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/78b593861f7849eea8f725b5e7187a34 : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":96},{"id":"7a69008395700d90fe6ce776bd77e1852612d14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:49:57.000000000","message":"Patch Set 96: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed2688fc6134512b9c0e3ff119849eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654b1211f0b4492b88a5004fc1d24372 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f88ed9bcc5d148de993c9fcd0982acec : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca683e31acee4b41a645b8ea20dae6b9 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb1efedd8824c30997337970702323c : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/15d00faf2812445dabc63b44729b49f2 : SUCCESS in 44m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a78acecbc5564e84abd53a4f67b64c65 : SUCCESS in 1h 03m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a4b56c4c85441c2b58f3197ba52f545 : SUCCESS in 1h 38m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d0ebe8a14054a81af84b345d62bbaf2 : SUCCESS in 35m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8225335e3848f2a47a5361235333a2 : SUCCESS in 15m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f499b0d93d864420a4bd850f1142dc3a : SUCCESS in 25m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e35821f74e12444c952af228aef7a5d6 : SUCCESS in 2m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/56f42a997326425aa17aa9f0cbd94414 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b888f2d452bf4171bf697bc9ee29e7ae : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a5ceb163bcfb41b6814a43fd25a58c31 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/289138daba584e9fa351262a54b08ab4 : FAILURE in 33m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/914177d537e64773b1b426bc58a6766a : SUCCESS in 3m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5afd3a40268a4b54a59609f38f4a761b : SUCCESS in 1h 13m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/db8d18cd63644b11b662962b15aa4b4c : FAILURE in 3h 03m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0924af69428445feb19f09546c4e3261 : SUCCESS in 1h 52m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/206b9b8e138740ef85afb6d5b36dfa5f : SUCCESS in 1h 54m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/71f30540f6634031a9e7598cc39e3495 : SUCCESS in 36m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3bd0322b05364d7e887c86f74726b0c0 : SUCCESS in 40m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f96ceb8859674e26a76a8c22b292cc6e : SUCCESS in 1h 26m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/65fd722a06b84c3687e2b1ddbe1e609a : SUCCESS in 26m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6441c5742e4545de807b2c352d30bb57 : SUCCESS in 1h 54m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2b536747b6fa42e2bd149a2d3f172cde : SUCCESS in 2h 02m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4e587116a5a46e5aa49eb144ce46d5f : SUCCESS in 1h 43m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b737da0e982948e098ff9b5e8787dc96 : SUCCESS in 33m 23s","accounts_in_message":[],"_revision_number":96},{"id":"25f6b2f3ef731ccd677f586e6f5f00c82b140c70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-28 13:56:50.000000000","message":"Patch Set 96:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_01_2026_13_50_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":96},{"id":"073fd6c0917617a20f887323b0dd7960f31226e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:43:15.000000000","message":"Patch Set 96:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebd7c0056ccb474a9252c47b0dfb547b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58b2a3d6df804d438633cafe4886b7ed : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b623779ca9124556b0803113e98407eb : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51865c71d8194e3b8efd438732fa9782 : RETRY_LIMIT in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":96},{"id":"51cea318fb6ab87bab1deaace340c58ac7ed48af","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-28 21:40:13.000000000","message":"Patch Set 96:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 59s","accounts_in_message":[],"_revision_number":96},{"id":"4f8375842b8bb1caaa1b4492fe97f2b06f7491d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-28 23:16:49.000000000","message":"Patch Set 96:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5947c947f2474bda920647264fa9753b : SUCCESS in 43m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ab9d3ab863cc4778a323bbce9010132c : RETRY_LIMIT in 15m 04s","accounts_in_message":[],"_revision_number":96},{"id":"14b89faf08f066ebc17afaadaf66eba1961b3a32","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-29 01:37:35.000000000","message":"Patch Set 95:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_374/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":95},{"id":"fa445d2d48a7312a08beda0bd7e39e6d48a4b3c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-29 11:51:22.000000000","message":"Uploaded patch set 97: New patch set was added with same tree, parent tree, and commit message as Patch Set 96.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":97},{"id":"5a46d9d4e7b028294a556fe7a2b7676a9fd14a80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:09.000000000","message":"Patch Set 97:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/727/","accounts_in_message":[],"_revision_number":97},{"id":"20bdfe2193f78bd26739d7a3d0c10e7ff739de35","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:17.000000000","message":"Patch Set 97:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/97/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/97/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":97},{"id":"d2375f40990d4fa214b22311f1e9c0f0bf35cb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:15:19.000000000","message":"Patch Set 97: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e645a004e3814d85b82666f70bf69b31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ef3c8a4353444b87994749d371e36e : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fc2c0d4060c446d9659b614f6462c41 : SUCCESS in 8m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a684c332c16c4e7e855bd602a7852930 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebccf58794b48a8a8d0dc0061496b0a : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/52001b01dbc9434983c21a35fd10d67b : SUCCESS in 50m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59fe514085ef42df87764cccd4bd1fc7 : SUCCESS in 1h 05m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7f4d8d3cb144f2e8949c9cfe38ad982 : SUCCESS in 1h 15m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e76690b84b64bb99653464eed482948 : SUCCESS in 33m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b5ffbe385db4b01a48b86a0ee1fa061 : SUCCESS in 14m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c18bd2e30bac4b3dbf7ec02b43274fe4 : FAILURE in 24m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/32ecf239f0de4efaa9b674d24e05cf68 : SUCCESS in 3m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6179769d797497ca6afbd5d26db56c0 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5217b0e83e4a430ab5220dd884396785 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/da4b5bb573ec4e07abf2d05282706b9c : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/801b844bd67443f7b806cb8b1c3ec16c : FAILURE in 24m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c97a07420b446d19ade0e6c25389574 : SUCCESS in 3m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61082ee04584454c8b2fa0999a9291f4 : SUCCESS in 1h 58m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5cfaa7a15bf14ccc97e7790f9828114d : FAILURE in 14m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14e1b8cba3754cf5ab10d0b57d46a4b7 : SUCCESS in 2h 01m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/770751d0cdfe4211bedc91d14aa40418 : SUCCESS in 1h 44m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9639f4aa4d2d4e4e92a108e7143e0ae0 : SUCCESS in 37m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/209481a7c9f043059addddcf663ca3ad : SUCCESS in 1h 24m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/527a6d2b6c2849d49466b2e755b13495 : SUCCESS in 1h 24m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c737b1b250614388a0cb1a05388d7b27 : SUCCESS in 37m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/08897c6ac514443c9ac7746801b5450a : SUCCESS in 2h 09m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/20d04e21d2eb47caa1ea51e2729a0d29 : SUCCESS in 2h 08m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a8110d9d6ee43b4a7a0d0d87ff0bfdb : SUCCESS in 1h 23m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e937b2e67238420196ccf9f76acce526 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":97},{"id":"76c775237c8b9528aa1a28288f1e91f159f90142","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-29 14:17:22.000000000","message":"Patch Set 97:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6d26572961e4486d8069d1d4060821bf : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":97},{"id":"8055961172edd2dca0ed638399d3fdcf5cc5dabc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-29 14:18:57.000000000","message":"Patch Set 97:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 44m 05s","accounts_in_message":[],"_revision_number":97},{"id":"ecdfad6c6bf5a51d8845ae78dee340a5f23e35ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-29 14:22:18.000000000","message":"Patch Set 97:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_01_2026_14_15_37\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":97},{"id":"0e2e98b6b140eff8a0ff6375cad50d5d8e1096ac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 14:27:06.000000000","message":"Patch Set 97:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":97},{"id":"83c165add3dab5e02a7d6f0d979b6430507d5c23","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-29 16:21:39.000000000","message":"Patch Set 97:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f27eb3596a664a99b849f6aa82d36e31 : SUCCESS in 52m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d8a8384dcb744ca928048c4706cf393 : RETRY_LIMIT in 15m 22s","accounts_in_message":[],"_revision_number":97},{"id":"811ba2f052dc70d05617445780b3b3f7d56d94a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 23:13:13.000000000","message":"Patch Set 97:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508d67678f96485993d79cc67e738133\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18a1edb34f3b440588dfa4388dfb278f : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e95412d524de4c56a96af5696ae02349 : SUCCESS in 23m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d973d70913de412ab0ab99cd1de6b47b : RETRY_LIMIT in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":97},{"id":"b85b87dfef3ed20761cd4665d021349b9b3860a7","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-30 03:17:17.000000000","message":"Patch Set 96:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_397/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":96},{"id":"8cfd29f18d24a0512f5a34715a03ed5db7ef952e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-01 03:10:19.000000000","message":"Patch Set 97:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_418/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":97},{"id":"bc760d0f73fb0a69a7f04e5da6f922cc0aa21462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-05 06:38:56.000000000","message":"Uploaded patch set 98: Patch Set 97 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":98},{"id":"bd34eb94f276248d6034c0d54e5b76a35a3f16cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-05 06:52:25.000000000","message":"Uploaded patch set 99: Patch Set 98 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":99},{"id":"4d5466c4627c0b4440d4ef90469cc53582dbca60","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:18:22.000000000","message":"Patch Set 99:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":99},{"id":"469a95539b6f57fe49cb01b8e96e6d46e0ae1ca6","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-05 07:18:41.000000000","message":"Patch Set 99:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":99},{"id":"09de1cb15bc3a061d8a64c3d3aa539a8a80b6a62","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 07:18:47.000000000","message":"Patch Set 99:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":99},{"id":"89e2e9f4cbae9eabebcbaa0f71fb43a906c999d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:40:25.000000000","message":"Patch Set 98:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":98},{"id":"e3c7db91f91b398ec5057332a2a2783e68e9f742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 08:24:45.000000000","message":"Patch Set 99:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e80381bcafea41b284cbfbfe83ecec5b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30b64c1c8c3a463fb817f6dc1365b68e : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77323d178ffe4f95b749492df33a6bd0 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/499ef1a9e1284365834e161b153926a0 : RETRY_LIMIT in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":99},{"id":"1f1cf27d65b63a4f291c93461799a4710078ce5b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 08:38:18.000000000","message":"Patch Set 99:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":99},{"id":"e7b03cc83fc6ada0e46cd7be11f07e1913d33e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 09:44:17.000000000","message":"Uploaded patch set 100: Patch Set 99 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":100},{"id":"4ddcc49679e70a6d7ad2e80ccc06d0bdffb52e60","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 09:53:50.000000000","message":"Patch Set 100:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":100},{"id":"0cc0891b8f74394a07586357e84884107af38504","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 10:25:04.000000000","message":"Patch Set 100:\n\n(2 comments)","accounts_in_message":[],"_revision_number":100},{"id":"705e61f2a568ad96b37ff23c90725922951c4292","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:43:12.000000000","message":"Patch Set 100:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d4c101a3244cebb0bd89eb306e72ce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a335805ab044dc5b573c17b35ca6919 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/423149726b9d48019ab3d13f1f6c9771 : SUCCESS in 14m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/095b281457ef49ce98c06584d470e41f : RETRY_LIMIT in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":100},{"id":"ed13a7fad0b11a75c0dcbaf903bd19b51034227d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 10:43:25.000000000","message":"Patch Set 99:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":99},{"id":"6c1ac7c5b9aa36d7a2992a5d70ca4d5de0dd7c59","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 11:54:25.000000000","message":"Patch Set 100:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/910f1802406e42dfb3e66dc57659df6f : FAILURE in 35m 04s","accounts_in_message":[],"_revision_number":100},{"id":"bbe00c72fa05071ffa7b8a736cfe70f858ab529b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:28:33.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2236dabac843c7b914e4792870bc62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fdc7fa94cc44089af6dd7ee5147c89 : SUCCESS in 9m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/668efa0c93194912bbc0805360bf8967 : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/268766acc4b442a29c6ef865178b0771 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d543aebec9a451c866e434d011ce7d4 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/aaa245909dd642b8a814747e170fa9e7 : SUCCESS in 1h 02m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3cb3149dc844e9aaa824a40dec9e292 : SUCCESS in 59m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0bec00dca854588a09e034cc38b08b2 : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2feeedf0a5145739a0725e113e961f3 : SUCCESS in 38m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38ea44ab6134218a9e9d9df4b5edc9a : SUCCESS in 18m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8317de2f19a54616ada97e56c6bc3c39 : TIMED_OUT in 41m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9eacb11f23be418e875c11dff15bbef2 : SUCCESS in 3m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4689b6ddcdcb4429983f5a0db8b7c28e : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f38320a1d77d4017ae94483c5d47b49e : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bb2c176328ca454e9be94379b27c1cdc : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67b97c9de82345eabebf82543a85874f : FAILURE in 32m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/233c0a2b3bc84d07bc4ac87ef58b5509 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cbfb4040c4444b139c23494ee90f8508 : SUCCESS in 1h 53m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e944642f781c45cca4786be442d3d488 : FAILURE in 2h 22m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0db0ee4d38e1452e83b8b744e640c2f0 : SUCCESS in 1h 52m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/05a625076d464afb9fbe0ea3d992f2f4 : SUCCESS in 1h 57m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/75dde609ec7f458d98ad5fabde8a3905 : SUCCESS in 37m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/283a8718e98b4c5092260beb7cd36383 : SUCCESS in 1h 15m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8558262f5b24fc39c2b12ba8ae37d78 : SUCCESS in 1h 33m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e89b1112a29423f9bbf0d3b3266d6c2 : SUCCESS in 29m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/09c4cf278c0744e89f4ea9b02de8d886 : SUCCESS in 1h 03m 49s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/913fbef73dc14dd182d612d150c1a689 : SUCCESS in 1h 14m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d47e415809724ba9a0456b207452ba2a : SUCCESS in 56m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01e7c1fee0054325a4bf5bfbda7f46d8 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":100},{"id":"12f1e8742d959a1cdce4dcdd61643cf92fb97187","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 12:54:45.000000000","message":"Patch Set 100:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_12_47_59\nLeave a comment with","accounts_in_message":[],"_revision_number":100},{"id":"381c367b1813532d510738f336f019f46195283a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 13:18:53.000000000","message":"Patch Set 100:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 36m 02s","accounts_in_message":[],"_revision_number":100},{"id":"5afb67488dc9c5459a0d01888f42165adea04d9c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 14:02:55.000000000","message":"Patch Set 100:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":100},{"id":"dc152f7af8ac3a8ea09e89cc0ff1455e19747056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-06 15:01:12.000000000","message":"Patch Set 101: Published edit on patch set 100.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":101},{"id":"ca4dd4a408da46ecd1123357eada11bc2eeed340","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-06 15:02:04.000000000","message":"Patch Set 101:\n\n(2 comments)","accounts_in_message":[],"_revision_number":101},{"id":"5d23b464fe7a0d1e7d68ab5d7a8377f206fbe432","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-06 15:14:37.000000000","message":"Patch Set 101:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":101},{"id":"36ec11c4b3d8f9556d94f7a0e5e750c2633ffd52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 15:23:40.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcfe409a8835449e91e104f323d7d081\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eada14a9b390401f92d7e92cc1ce124c : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b726622a1b0483ebf289cba1784ae53 : SUCCESS in 17m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a25512dce73437da555140e0556729c : RETRY_LIMIT in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"20377b6ad2471be68005a0d4a21f0d61356ac4f2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-06 15:42:30.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 21s","accounts_in_message":[],"_revision_number":101},{"id":"1342b444c86b6f0b26069bedecd1edc6e207ee50","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-06 15:48:25.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bcbc9279eb514c9481914e15b4a62edd : FAILURE in 44m 04s","accounts_in_message":[],"_revision_number":101},{"id":"f926fbbd3f0c1e5cb5948a03e971af54c3148108","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-06 17:19:07.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9d45d62d210e4ff7a985f27eb093b87d : FAILURE in 58m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8620e32d24414972bfad1a8c7a8a8d0b : FAILURE in 1h 16m 58s","accounts_in_message":[],"_revision_number":101},{"id":"7ff6467a090e07119ebac015f5a35f343cc344c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 17:32:21.000000000","message":"Patch Set 101: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea44113df614870abce5938dbbf73f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcefa693f2e2490bb9c23090dda78648 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22e7337c887c4c6baca4892eab6d0684 : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be53ad89de7e4fbfa41527d486417679 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1dc92a36814e16a745473e99809d46 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/a83cc74bfdc84339ab89314abe8a63c6 : SUCCESS in 1h 03m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b4b9bb9cd424dffa30c074686a503c7 : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe0b9b6bd11d491d81bd36adc307a347 : SUCCESS in 1h 52m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8107b658504494ba4df43c939b97565 : FAILURE in 1h 05m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a47226aa9764f81928588e7c3cf26b2 : SUCCESS in 13m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9bee189714c47f08d847c26ff689c2b : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/457a46f095024e079e39e7fbe686c416 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38c3aa61f1484bd7a2f40829d3a4c245 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb726453345748188226034ad3b51bdb : SUCCESS in 5m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/33c638b2649848b498c91e0134f8658c : SUCCESS in 8m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e7942bf8b064f389df4254cf5c32f8a : FAILURE in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1679dd9d76743d38a5838aa2a7877d9 : SUCCESS in 3m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0fabdff9b8147919abac42213f32908 : SUCCESS in 1h 58m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/edf5434a70c44cc1a10a18460dcd5262 : SUCCESS in 2h 24m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/afe6f93358854b148a06a277f8630dbb : FAILURE in 1h 48m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f2bf861e36bb49a1b7578e537a152a83 : SUCCESS in 2h 01m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3a835a6b66884088b597804f01e022a8 : SUCCESS in 17m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91293b3e97554b31a95f0055aba23f56 : SUCCESS in 36m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91e05df2d1814fe089039a90223a35cb : SUCCESS in 1h 33m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e2ce864d9914a0d96e4b0d7ee6f185e : SUCCESS in 46m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/846d806002834749a243bac4c3cc7dc5 : SUCCESS in 2h 00m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1429120a749f40ee87c63185db49bfea : SUCCESS in 1h 03m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55b6e3b2b3d84d959853bd43b5f204c7 : SUCCESS in 1h 28m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbdfb9d10b2941b2bbe7699ddd04d799 : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":101},{"id":"d80a885625c3e72217fd1c493618f9663f42c2de","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-11 09:09:40.000000000","message":"Patch Set 101:\n\n(1 comment)","accounts_in_message":[],"_revision_number":101},{"id":"d3d966a00f94be069552b78f6b5cb9d843633985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 09:32:02.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3ec341332b43fc91524028b26fafe2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/533252d869ba450bbca19daa66305e44 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b664a00b8c84b8cb723ca6c5056399f : FAILURE in 12m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/508cf228c41b41c986981b727cbaa37d : RETRY_LIMIT in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"2f2a6949b29599ebbc3911a5975b2f94bce0d5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:22:49.000000000","message":"Patch Set 101:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f8f7ca6054741a4b80d630123ef50d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c505e1318ae74c0986c712c764224da8 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dacbaf53139046baa29262f6594141de : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/817bcf1054c9464ca6d795ee9df06e41 : FAILURE in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41bf7b81a6343d8ac2f4d30e470e26b : FAILURE in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e8fb6af72c1b45b79771c250d19758b8 : SUCCESS in 1h 01m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8cf1305b38a94c23ac9efddb5372db60 : SUCCESS in 57m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fddfec8bd9942b39f1ce394b94aee0c : SUCCESS in 1h 48m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ef8d7bfb6ff477094c6720fd5520677 : FAILURE in 54m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e02dde620e214641a513042f41f3459c : SUCCESS in 19m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4ea5f8ca81b84e11ab21f798a850f8d4 : FAILURE in 5m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48ea3d1b900b4abc9a4ab7652ec2ffff : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87207817699a40419b22138d2bd002c9 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0dc6a02fdf1040829d8d31e5c3ad6c73 : SUCCESS in 3m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cabe019c59d04e81a1fb4ca07b01d31e : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09966e86fc7b4067a6d6ae5a1683e1fe : FAILURE in 41m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/332ac8490b1a4584bd429c95862b3faf : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/609cf18269de4d24965dec508529fba7 : SUCCESS in 2h 01m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c43c21d04ba84431b7a6d43ec8945107 : FAILURE in 2h 59m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/866ba624651a42e3be4b8db935a69ac2 : SUCCESS in 1h 54m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/45fa42a44bd74c64b2e613b9f404545f : SUCCESS in 2h 00m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/920f29cc8fed40268cd13ad1ffe30566 : SUCCESS in 37m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e8a48f6383c4a2ca0af8985094939ec : SUCCESS in 50m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9229c9e400b40198890d094006f1607 : SUCCESS in 1h 34m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bbb5608354ef4b8dbd329e2e17122efa : SUCCESS in 27m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/127a8b8987da4dcdb4f8322cf853cbb7 : SUCCESS in 2h 28m 39s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/af8cc6e4475f429b938b06b1f81ea00e : SUCCESS in 2h 01m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/151f0dd5060d442ea99daf1fde14a375 : SUCCESS in 2h 00m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/553e9b328e504f03ad63391750b25bd4 : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":101},{"id":"a4c5e98f9bfb2798ad8c8130c4f9c7d17ffc8079","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-02-11 14:59:58.000000000","message":"Patch Set 101: Code-Review+1","accounts_in_message":[],"_revision_number":101},{"id":"482d0f06f0938d7b964fffc4d0a4616e2193be07","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-11 15:51:04.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aaeec72b194c4627b31b931301e42314 : FAILURE in 52m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/921b94708f55448b8134b07f919bc328 : FAILURE in 1h 00m 29s","accounts_in_message":[],"_revision_number":101},{"id":"a0dfea9251d864620fe5df9a5e49599242b6599a","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-12 07:49:26.000000000","message":"Patch Set 101:\n\n(1 comment)","accounts_in_message":[],"_revision_number":101},{"id":"33eb44ed14c05bbf830535c5357bdbbd5d39b8c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:25:44.000000000","message":"Patch Set 101:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3993de8fcca64f51a6fe6b7a924e5a18\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0758e82ee89941e88993e8040474ae4f : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e6a3e780d3c4d9a8b657c16096b8405 : FAILURE in 9m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9a1cff754cf4d6bbe7b4479edfc22da : RETRY_LIMIT in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":101},{"id":"10832960d400de6668b790baec266fc3efcdd1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:59:28.000000000","message":"Patch Set 101:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5f234e8757749549c094dcc55b34aab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5774d55613496886a502b6c09c00f1 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6bceece5515483880fbd4a1c639cce1 : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77b54769e7df46e8bc88ed04e8761039 : FAILURE in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f273b4319346e6a868f2ee7f52a363 : FAILURE in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/7bacf487a9bb490eb77e05b14c9b375e : SUCCESS in 58m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92f91a9d2e764da7831d12493df63610 : SUCCESS in 55m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3ec275ced7c423d975814da05cc0483 : SUCCESS in 1h 37m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c77cb1b57074d5ca93764444b89d2a6 : SUCCESS in 59m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64190f8af9e6455abaa54a9295fa7f5c : SUCCESS in 16m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8e3066c8fe3948ac89708723e8ad40f9 : FAILURE in 4m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fcb4270190b4cb2acfb0d0578d6b1c3 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/352f6b6b6ecd47bc8de6a35823f9605a : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb3aa6fc5eea4b229b2ef545a4a2a94d : SUCCESS in 4m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/78569e48d8a9424697354341a047eb3c : SUCCESS in 4m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b14cef8469847e0b1cb24052ee00419 : FAILURE in 44m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c32eb8638a2499cafa357f68b8a5c56 : SUCCESS in 3m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/985458438762452fb404fed6ea5c9ac2 : SUCCESS in 2h 11m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ef0bb69d9aac427baed4d04c90dba90c : TIMED_OUT in 3h 05m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/56608d4df24b41b392d135dd6c09329c : SUCCESS in 1h 51m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/47ae58eee0de4c0580907648d5a52003 : SUCCESS in 1h 52m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e27fc5f366a141648eb5bbda5f952044 : SUCCESS in 37m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c81b073f82fd455da44f32da8c9d648e : SUCCESS in 1h 05m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91e7d94911414c718c56b3c33759360c : SUCCESS in 1h 28m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b56723aeb95489fa0936c1d0b0a80e6 : SUCCESS in 38m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28ffd0cdab8e4706859bb2b87bc36553 : SUCCESS in 1h 51m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/42fccf1eaf824c0382b675d557d3b044 : SUCCESS in 2h 11m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/adcbd3ff9cfa4dbab01249ea2986846d : SUCCESS in 1h 42m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a71178b975104618932f4e2331da2212 : POST_FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":101},{"id":"f76394bb1c40f1733d3135361d7985eda3779260","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-13 02:03:34.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1dd2873330bd4232b1528d260832b18b : FAILURE in 1h 00m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e33ec71ac4b9499a9057d654e8c66f71 : RETRY_LIMIT in 10m 45s","accounts_in_message":[],"_revision_number":101},{"id":"65d3bdcff1a510df61914bc84ce30bf83c9b24ba","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:32:14.000000000","message":"Patch Set 101:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/dee82c7d12f84a3692d529b03afdb283 : NODE_FAILURE Node request 200-0000623372 failed in 0s","accounts_in_message":[],"_revision_number":101},{"id":"8deab0effa8ecdf8a5c017e3a9f4a889b21bb0c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-13 09:22:24.000000000","message":"Uploaded patch set 102: Patch Set 101 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":102},{"id":"1310cbdccca5dadbcc5680568acb2b1e32dd6ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:58:58.000000000","message":"Patch Set 102:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c4a4aaccf143dcb898bff355f2fd93\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/814704973a61484b92c1e4edae31ebff : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da1a4de590234e38ae85804675e075a5 : FAILURE in 8m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1d939bac67304df7b9ffea23caff4f27 : RETRY_LIMIT in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":102},{"id":"ec9a95fb128a9537362831f50d8ffce731c89719","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 10:00:29.000000000","message":"Patch Set 102:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":102},{"id":"c41258fe2288424e29313a0106649c336775f614","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 10:57:04.000000000","message":"Patch Set 102:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d2e9b863a3f14bcea0a9537ce2fd5a21 : FAILURE in 23m 44s","accounts_in_message":[],"_revision_number":102},{"id":"2e5a2d644e8d8ea3b0a77470567763e1018c9709","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-13 11:43:47.000000000","message":"Patch Set 102:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 35m 15s","accounts_in_message":[],"_revision_number":102},{"id":"abb10708f294da415a9162d16b74680b99815dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:01:16.000000000","message":"Patch Set 102: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa47f512563041eb8e7b73e0cca8c9d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbe4409202440f6b1bee56d61f2fd4f : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5905b29e8ac94350987c545e47b95ea4 : SUCCESS in 7m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5779f36cc898453ab276479f00b087bf : FAILURE in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c306c6954a4a9786415a3f1fb06834 : FAILURE in 4m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d5cd43f0051e496996cecfe4cdebcf28 : SUCCESS in 53m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dbf72009119f40d0ae2a59e8a72bf9d1 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f980c42ebfb2489b9854ec40a34dacca : SUCCESS in 1h 24m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22d2a6a0d8894cf1be8d5b9e6da7568c : SUCCESS in 1h 01m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc4734c14de4d8096f803ea2aaf74b7 : SUCCESS in 12m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34fffb02adf24388863149a56fd0f383 : FAILURE in 5m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d0b35aedacb94687a383c62e9929367b : SUCCESS in 3m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3158152b3867474b9c4d03627a49f43c : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1113248d83cc48c5843bdbc365b019f2 : SUCCESS in 4m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b92d3797acc24a2fb297e050469b2bdd : SUCCESS in 4m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc0e95194fa8489aa950075795b60219 : FAILURE in 48m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d3127d34ed7456a83b34333ee4cf25b : SUCCESS in 3m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cecc5b61bda04e26bc019819b0d082b3 : SUCCESS in 1h 57m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8a8682bac4cb4b2697899cbff23f9880 : FAILURE in 2h 26m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4a5fd83b3f24d35b8cd766ac0ee558b : FAILURE in 1h 38m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/51d4fb718ea7467591351e7ed6aa1219 : SUCCESS in 2h 04m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a468b40b2d1440299fce3e13de9a0ccf : SUCCESS in 38m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ec9e3d4ba594cdea6b303294b6b1a5a : SUCCESS in 1h 09m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a303bf44e794841810062fac3517a49 : SUCCESS in 1h 43m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/847f2bf6078a4cccb61c4f74cb158c57 : SUCCESS in 40m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3056a55c35a34d208d160add04564ac1 : SUCCESS in 1h 39m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/19cf6601877a4b1bbff5fe0f35c34c01 : FAILURE in 2h 25m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/722423b9a5a84c3480e6215482c71469 : SUCCESS in 47m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2ddd1fe611a45248587bede26021b32 : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":102},{"id":"14a094fd239170748378b2856f190c313d42dfdd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-13 12:14:39.000000000","message":"Patch Set 102:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d2cd1901cdba42bf8da10a014ed895c7 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":102},{"id":"1856f7e0b862c74df4fd790d0f6368ea6e0d1c50","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-13 14:28:42.000000000","message":"Patch Set 102:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_638/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":102},{"id":"4e46e3aa8411d1c2a8756c52ce555c075056921b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-14 03:14:51.000000000","message":"Patch Set 102:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c6a16e7cc0cf48e1b7b82a928768e848 : FAILURE in 58m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5b4a86b8793e4530abf12f3959bf0a4d : FAILURE in 1h 00m 07s","accounts_in_message":[],"_revision_number":102},{"id":"69ffd7bfc6b8e0852195485c8ae3f286da424bb4","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-17 07:53:05.000000000","message":"Patch Set 102:\n\n(1 comment)","accounts_in_message":[],"_revision_number":102},{"id":"d65374fc13676ad593c3717d8b8b54e76ba670eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 08:17:28.000000000","message":"Patch Set 102:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9545a631c54cd1b3b6ecc7038808b4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92803b31fe9a4297b8cd7c3db108f3be : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/819d296451e748ca9d2f10a2a274fe83 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68321c0943b840d78c7a4de3612039f8 : RETRY_LIMIT in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":102},{"id":"e8d75273308573da92c77900f557a40c86738a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 10:18:33.000000000","message":"Patch Set 102: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0738753592e64d15a06b5db644bcd49d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a62a9bdb5d14e9e8a027fcf623844ae : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d6cba17b1084ac291d84303732bc598 : SUCCESS in 10m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/132ba656249e466ca88456862b07a057 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b8fdc7dac04975808f1a99484c683b : SUCCESS in 15m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/d0c6b856f685483fbcd249dbea1e252f : SUCCESS in 1h 01m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e2987a708fc4f7e99b95487d1ebe30a : SUCCESS in 59m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ca31b94a8694d4590041de0204d3f0d : SUCCESS in 1h 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c992a9f758464fdb80bb94ddbe2e8147 : SUCCESS in 59m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/954f0d15d91b4ace964e52deee58f95f : SUCCESS in 15m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c69be086b52441d88bf5b01987ecaf75 : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/815baca87dec4b7f811fad4a06f6905d : SUCCESS in 6m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b4258988dbc4a118a32617cdca9778a : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fef7794e81dc4d42867668a5cb6eb4c1 : SUCCESS in 4m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/32e96e92ab394c2e93e476eb189acaca : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1626fae939404a15b55804b2e724c2e9 : FAILURE in 27m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cd311696ae348faa2f73d00c9e2d265 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1703e08c46964c8ab6e273fe84d2c964 : SUCCESS in 1h 46m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2bd8f3e22c74b78962a78ba2dfe1124 : SUCCESS in 2h 11m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/12f9873d605c423e8efd21a15d95faa6 : SUCCESS in 1h 37m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fce6496e84144641a4be06725efde550 : SUCCESS in 2h 04m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ecd975e6a6648a98b59fae8757a8315 : SUCCESS in 39m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0fa1c41bf654f5183ece0dfc9cbf66b : SUCCESS in 1h 10m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/12dcf39f685d4ecf86b356cfc0b5ab0d : SUCCESS in 1h 55m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2664502e4e1e4aa58764b32dee304371 : SUCCESS in 44m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f2cb0b6646c249b4936f6c2d79440244 : SUCCESS in 2h 02m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/71ba4f94d7d14c71b37c6f7e8e2774db : SUCCESS in 1h 49m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/014557866cc9410e9c7018eeece6bb9c : SUCCESS in 46m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/284d984d7b8b491a83af193f9182b98d : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":102},{"id":"f1a6dde6d32bf9e13f73204f5abc42b0b46d1b71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-17 10:35:54.000000000","message":"Patch Set 102:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_02_2026_10_29_06\nLeave a comment with","accounts_in_message":[],"_revision_number":102},{"id":"08840cd6fb1cedd0cad5e3b9bb26053ec8d8f04a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-17 16:40:25.000000000","message":"Patch Set 102:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ff7d60a8bb4d494f808308c2bb68f0f1 : FAILURE in 14m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/24944bf9c843400282a52cfcd4459020 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":102},{"id":"936d0471aa6730668587aad17f6146026e374b1a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 11:24:24.000000000","message":"Patch Set 102:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/800/","accounts_in_message":[],"_revision_number":102},{"id":"6e81c1a2f16b27e98d5e6b0e720cfbf7f0a045c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 11:25:16.000000000","message":"Patch Set 102:\n\nBuild Aborted \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/102/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/102/fc.log : ABORTED","accounts_in_message":[],"_revision_number":102},{"id":"0894fb610ec9787478245cfbef37d9c34a67accb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 11:25:20.000000000","message":"Patch Set 102:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/801/","accounts_in_message":[],"_revision_number":102},{"id":"cd040d6c365c0f32e364c0b3f976e2d571cca326","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 11:32:22.000000000","message":"Patch Set 102:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/102/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/938771/102/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":102},{"id":"ea144c89ffdabea04f012936df72d106cecae500","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-02-19 12:24:48.000000000","message":"Uploaded patch set 103: Patch Set 102 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":103},{"id":"d4d1630228e31f845411ae24d1c6981c1d3ceabd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 12:42:21.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eb64f7b67624447b8692439066aa2943 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b4bba2e14ea1450dbd7d5731a081c009 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":103},{"id":"385197b3efbfe410dba90b9d78bda57e9af7e6b3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 13:08:08.000000000","message":"Patch Set 103:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":103},{"id":"4ac631f26c5115a8d9e7791899665fb16abf398a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:28:22.000000000","message":"Patch Set 103:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d55a3d99440e4715916dbc987a07b4e3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bbc3d174d4b49c39cd883de0470d62c : POST_FAILURE in 11m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/029205d7a79143939a15a4eac2f5645e : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a323cd115a174ca98d8a97077e3a7fb4 : RETRY_LIMIT in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":103},{"id":"b6df9a5f186c8516fd938ea09cb000b0a007e6b1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-19 14:42:59.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- hpe-fc-driver build/213140270dfc4d5d801f5ee064be93d0 : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":103},{"id":"178e880e7ce83914e2e658975251f011ca7dfc71","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-19 14:43:31.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/1f62efa334174aa7af8457038de487a9 : FAILURE in 45m 11s","accounts_in_message":[],"_revision_number":103},{"id":"bb256676437cefe4788c88c86e22a8c67a51cf80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:46:09.000000000","message":"Patch Set 103: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3590680e77474818803dd9faa1bea141\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01395a8faa14335ba756e401c22d806 : SUCCESS in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0e6e046c6e428c9aa834360e51b350 : POST_FAILURE in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c3e4ef16851492193f74e33ec6b927d : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa90212ea7742bc8f3a81d70d161587 : SUCCESS in 15m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/4dee8d1018b04853b270f53926b0a3f4 : POST_FAILURE in 1h 07m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/087e0e4dafe249c899196ff4d1829641 : POST_FAILURE in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/974b442d00e747168371d2b1b3d4d692 : SUCCESS in 1h 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a0f1d550f46466c8ba8c41f46ae1a4c : SUCCESS in 57m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0af47979c5b486a8a054befe0287495 : SUCCESS in 19m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/adc4ef99f2f148f2b82a460c7116a006 : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ecf0b2911164bd2b769c7a159e3e818 : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b45764ffc1c4f5f88bb363e7af71b6a : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb20711c03384599adcbc8ab4bb83108 : POST_FAILURE in 5m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eee9bba9eb924f90be58357e5be64643 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a32499e7ad834820b77225531cf1f793 : FAILURE in 45m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01d6801bc64043c8a21173606661a229 : FAILURE in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e71768b95ffb4cdd9fcedf0ab7af6aeb : SUCCESS in 1h 56m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cdfd08938b73450aae5910ac0801b317 : FAILURE in 2h 12m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a048320270f24bf18c5f54767d28a8a7 : SUCCESS in 2h 12m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/59fd1c8ac5e841ebbc500475c65a608d : SUCCESS in 1h 40m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8dd1bdd58f464cbbb03c61467dcf1527 : SUCCESS in 42m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc282280dbc5452fa5370425203eca51 : POST_FAILURE in 1h 14m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7c66b09d51e54480ad1b085708e7e07a : SUCCESS in 1h 50m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2afce0ac85dc49298a00e7cc78345156 : SUCCESS in 53m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5ce987a3e7a14e909be7eca946ab1ced : SUCCESS in 2h 15m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/858136a2cdba4965ac4b0ccce3b4a06c : SUCCESS in 2h 10m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/329d47289797421facb1aeffa055d32c : SUCCESS in 1h 45m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8956f376cca4d959658980b0d531d1b : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":103},{"id":"4a098cdf460665791f99f681c00e2eb99c597663","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-19 15:02:07.000000000","message":"Patch Set 103:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/938771 : FAILURE","accounts_in_message":[],"_revision_number":103},{"id":"994b8120e2ec1fe2569c717a0e4ad27dfad1c4f2","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-19 15:13:07.000000000","message":"Patch Set 103:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : FAILURE","accounts_in_message":[],"_revision_number":103},{"id":"f4f40ae7b8de6b1700ff0942624c4a82ae2f19a2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 15:23:26.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/65a6acf08ed741f3a0cfa8fdb702b12c : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":103},{"id":"53b8579cf9ff8a8bb944332ad1554d3e453dfcfc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 15:27:53.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 59s","accounts_in_message":[],"_revision_number":103},{"id":"326677d722f076807938f81964c9e13017c8b24f","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-20 02:18:20.000000000","message":"Patch Set 103:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_764/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":103},{"id":"77a5848159910e761a7720d9f9d1c43d9f522dd6","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2026-02-20 10:48:42.000000000","message":"Patch Set 103:\n\n(1 comment)","accounts_in_message":[],"_revision_number":103},{"id":"f1e27e1222da8e5a4c0b9f258b16d4c2d630708a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 11:03:03.000000000","message":"Patch Set 103:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bdc2f132d09f42abb36f846425f0f213 : RETRY_LIMIT in 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1899f0699ebc4ac3a01875844aa39391 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":103},{"id":"79b5cd17b328c1ca5c98d05229e06399275d9b86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:12:57.000000000","message":"Patch Set 103:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9405e321bac4414d90dd72eb6d7127e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d8079d19b574e1bae938f3cd98b6275 : POST_FAILURE in 14m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07e39feab06e4fa3823bf66fe93369b3 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f6f0f45698194c75a074233beaf36373 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":103},{"id":"64182b5d1fb8715ba9d680ab5d80b81cf088b0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:11:27.000000000","message":"Patch Set 103: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1b5368afe2419ca1140fa342827717\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6199a53afd4647209d7645677ea62f90 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6a65834024a4adb97292da4f4ed9fb6 : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acb3a44eff0f4beb9a180bce6b6394d0 : SUCCESS in 12m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0756ab87a39a41a890f1c49f0265b197 : SUCCESS in 14m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162cc5e4143842b1b4a5e3fde3c9e8be : SUCCESS in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/3a59cf6e8ac546d6b8532eacc68046f2 : SUCCESS in 54m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52a07cc5ccfe439391675a78d719dd79 : SUCCESS in 55m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bd462600ac04ccb95ab256152600d32 : SUCCESS in 1h 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02a0eca497b846d883d2b1f45e0b377a : SUCCESS in 1h 00m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d58c78560414a829a643ab691e659d8 : SUCCESS in 13m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77d816339de64f3894ff0e63506e8e3c : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bdabb6419adf45c9898a423f09a641e6 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36b91ce45779482991fd680ec951632b : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98d55810c5d64f1c9bcd6cbec1f4fda4 : SUCCESS in 3m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e9585e13fe7846e0b2fef429ff290953 : SUCCESS in 3m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c270a9955af43d4948ab053bc5bb0fe : FAILURE in 40m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/678ca82f81734fc793c43b62c37d12e6 : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f21748476004acdb6c58899e8e2fb41 : SUCCESS in 1h 31m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3abc3fe7ece347ddb518e83e4ae7e213 : FAILURE in 2h 06m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1de88987f5744b82b59adfe189cb95ad : SUCCESS in 2h 00m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d45698763c9a4bf0beb2dd1635a7762b : SUCCESS in 1h 43m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cf915f7fab89426ca549aa9482eadb18 : SUCCESS in 35m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b3e7fff6ab144d7a2a41016a1a2788f : SUCCESS in 1h 03m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e878e707eca04850b2aea64730e5d1d4 : SUCCESS in 1h 24m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/50f89791f99248798961213a73c69806 : SUCCESS in 28m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/499290235d354178a54b4b8137087709 : SUCCESS in 59m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9a8be40e975240c2be811d95d127add3 : SUCCESS in 2h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb0b57760d014b6e8c490b73a685021a : SUCCESS in 1h 41m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6387c1ff7e814acda01d9c0ce1d3d241 : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":103},{"id":"0df93179e84f0a30e496f60ec34e5ac20c99fcce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-20 13:34:07.000000000","message":"Patch Set 103:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_02_2026_13_27_14\nLeave a comment with","accounts_in_message":[],"_revision_number":103},{"id":"fa92b719f8854ab3c733b6c522e76878fded0b26","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-20 15:50:51.000000000","message":"Patch Set 103: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":103},{"id":"7542581207104498a02c3ca528b204e54cadfc18","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2026-02-23 05:56:05.000000000","message":"Patch Set 103:\n\n(1 comment)","accounts_in_message":[],"_revision_number":103},{"id":"5cf39130ea0e639ee01e4f5b2df98a54be2b8f30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-23 08:23:16.000000000","message":"Uploaded patch set 104: Patch Set 103 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":104},{"id":"e44b1c787ba7a26c107d2b77f71461dbd5f1e7e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-23 08:38:23.000000000","message":"Patch Set 104:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2c234581415544acb5c3bc53c85068fd : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a798b1be66df4d489ddb564ed42a52cf : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":104},{"id":"1a18033ab64ebf3e06121df7ea65448a9ad78abb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 08:58:16.000000000","message":"Patch Set 104:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b8d3617cf74a86aa233c525a035f13\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e67771f917a408aabdc81ffded3ddd8 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f06ab28119954b34a07170a903a8a6a5 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c73017698bd48b9a1989bd204f7ad5a : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":104},{"id":"ac7a6a5c48af0e615c4c707e463c568d09af8f49","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-23 09:03:20.000000000","message":"Patch Set 104:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":104},{"id":"f55fd554990ce5798ed44a10d6f142043838b7a3","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-23 09:53:10.000000000","message":"Patch Set 104:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_815/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":104},{"id":"848663d24e0bb962a3b5f5a7da4ab9615c57c326","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-23 10:10:25.000000000","message":"Patch Set 104:\n\nBuild failed.\n\n- hpe-fc-driver build/3020a12cb0b1455d98e55bbddd833d25 : FAILURE in 27m 25s","accounts_in_message":[],"_revision_number":104},{"id":"73e725e3b048442364ced91a826a574a9c6a5c54","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-23 10:16:47.000000000","message":"Patch Set 104:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/938771 : FAILURE","accounts_in_message":[],"_revision_number":104},{"id":"c98c1078c624822f772ca1111ad573846fc902ed","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-23 10:28:12.000000000","message":"Patch Set 104:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : FAILURE","accounts_in_message":[],"_revision_number":104},{"id":"5408cf675b95e59a2b6b17bc1abed1287ad988a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:33:23.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13121ffbc4b94f58a5156749c19e3888\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f24b164f28431e82dda59c77c61758 : SUCCESS in 7m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21d13be9cd89480280d29e3c979f9e94 : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1443a0df83b44770a4b70bc2e168772b : SUCCESS in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d5f1441c222d4aa6b10d1881a1c86915 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fb9f75dca047a7b9903609d3d130e8 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a5fa6a42a071434ba46c31ae3c1d8dd4 : SUCCESS in 56m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/281545844b43433aa8eb9c5e123e27ae : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83050383e724477281d243d9d0ae4869 : SUCCESS in 1h 26m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2e585b01e6d47b189a4b18c635515df : SUCCESS in 1h 05m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5c176947fd94805b820d87399e52531 : SUCCESS in 16m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c42a885d90074ef39fdc05b1a257b411 : TIMED_OUT in 42m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be83f39f02e2423ba4a1cd2632aaf8f3 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca832618b02e46d194effc7227e25172 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/adf3607a4ede46b9aacc3e6c776e20f6 : SUCCESS in 5m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa00fd9c2c354e29b584e3e27ca731dc : SUCCESS in 5m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/085b4209c7ce447bbf272285d2f2cc29 : FAILURE in 45m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ffd2d661d2f346328dd819f964a198a7 : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/198af7ac4da3499bb66f6f7f7f155730 : SUCCESS in 1h 33m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6943578400014429b8bbb6b94a99fc3c : FAILURE in 1h 34m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92a3c7faee1c45c6a7ad6cb19bf147c4 : SUCCESS in 1h 47m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fd6e7685f1304fc39b09c6ae9049e190 : SUCCESS in 2h 01m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1ee40d36074462b92a29d4df73ced33 : SUCCESS in 29m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7757f09876c54c0a8d6618031c08e14c : SUCCESS in 1h 04m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2541857296434713bceb3437b81a4697 : SUCCESS in 1h 28m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b594c73e2be24c95bc74dfd8460517ca : SUCCESS in 34m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fe5c8eb4cca74f38b59ba0272333a7f4 : SUCCESS in 2h 02m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/50a9af5e92124b26b5872e3207b8298d : SUCCESS in 1h 37m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abd6ec4355ac47aea35c4bcc2f67ff36 : SUCCESS in 1h 43m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec42f3095e0a4569accfbff9139364e0 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":104},{"id":"33300b026499b97e46160df3205e72277956d202","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-23 10:39:23.000000000","message":"Patch Set 104:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5819c77f5b664e67bea0e14278bc893a : FAILURE in 35m 01s","accounts_in_message":[],"_revision_number":104},{"id":"b482138b5640020cf17e4365077e34f2707bf06e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-23 10:43:25.000000000","message":"Patch Set 104:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":104},{"id":"775c3639e026b53be9dfed158dd28aa78330a921","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-23 10:43:47.000000000","message":"Patch Set 104:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_02_2026_10_36_36\nLeave a comment with","accounts_in_message":[],"_revision_number":104},{"id":"28c118975ece29c513f3aa9cfc8be273d0934400","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-23 10:53:18.000000000","message":"Patch Set 104:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2096fcd33034400bb7e87f3dac687cd2 : FAILURE in 39m 37s","accounts_in_message":[],"_revision_number":104},{"id":"0c761ed2ef68854b4a9ad4e40cef174ad128289a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-25 15:07:25.000000000","message":"Patch Set 105: Published edit on patch set 104.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":105},{"id":"8865e55673a9e6f6006e197dbd4539972b0a710f","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-25 15:12:59.000000000","message":"Patch Set 105:\n\n(2 comments)","accounts_in_message":[],"_revision_number":105},{"id":"dca946bd918c0198b3f37044f7ab74dba237ea1a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-25 15:13:17.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/98a5ffbec480473cbcd15b86a3d33ee9 : RETRY_LIMIT in 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5e80a6cd2bf049e1b7dfbc88bf88e9fd : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":105},{"id":"f952cf2423db64906cdc5834eae0d7dd822d10ea","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-25 15:43:01.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- hpe-fc-driver build/db6a1bc0b7c24a84b4f5ee3ea0a433bc : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":105},{"id":"0baa3d3cbe1e1d7f4d7cebbaaca27c372ad8ff8f","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-25 15:45:35.000000000","message":"Patch Set 105:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":105},{"id":"17e0288c8f1fd1b17c2ff3fce72c90e054bea47c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-25 15:48:24.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 15s","accounts_in_message":[],"_revision_number":105},{"id":"7ed6ad1287a3bf8f86adf2127b864eeeec70a856","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-25 15:50:04.000000000","message":"Patch Set 105:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":105},{"id":"638a8bcfea07613632a690a4a620772b8034c66b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-25 15:51:22.000000000","message":"Patch Set 105:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":105},{"id":"fc332e83822d4e36391c110388cc37066dfd8ccf","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-25 15:52:22.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3d9232b6c0ac4d76be80789f4206660c : FAILURE in 44m 24s","accounts_in_message":[],"_revision_number":105},{"id":"bb91b117cd7624d8a74a8f393b5db129ce8e5bcd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-25 15:53:15.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/af6a7e73db0d4fac85143eba086b0cfa : FAILURE in 44m 48s","accounts_in_message":[],"_revision_number":105},{"id":"db65580f8debd1f43b2d6650c511434fa11e974d","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-25 16:33:47.000000000","message":"Patch Set 105:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_849/index.html : SUCCESS in 86m","accounts_in_message":[],"_revision_number":105},{"id":"c5063ece07b939be10e8dd968738867effbf5e99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 18:12:23.000000000","message":"Patch Set 105: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fea828e649041fbb657752ffdd4c0f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbbd7c9b5d9e49c78040379b21a90fb9 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/088e23766f1640d78514051bbf6ccab0 : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7512dd52e5174a93bf4d72209ec72b4e : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50291b4a1de5411ebd98bd9edccdf81d : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2aedfbf4bed491fab78ca7a48f045f0 : SUCCESS in 15m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/05f1edfcee064175be45e5c89d4a26f0 : SUCCESS in 1h 02m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72e379dc8470404b805d676dd05545f8 : SUCCESS in 1h 04m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89fc0f74ba63449686af83191ecd1dd7 : SUCCESS in 1h 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f37ea03c0d994ecbb56d8ec6ee180636 : SUCCESS in 1h 00m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/093ddfc05e6048efae8efc6942f069f6 : SUCCESS in 14m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08d8788526744e86b16d624dc5b4c64b : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae370b5c9e364dbc826bc9dfa2a1d02e : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1bf7b44898bc45e3a131c04f1d53eaf3 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e9ff98b56d3433daea0da79625a6f7e : SUCCESS in 4m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1f0714d44cbd498db9c226343c7f4244 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a391a89d2954fbc8eae2a8aec9978d3 : FAILURE in 39m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10d9bca065a44d8fa344426e1ef6e28a : FAILURE in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ccb58feeadbe4d148f8aa33772c5e251 : SUCCESS in 1h 52m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d9654570721b43e096fb1099fe2bd125 : FAILURE in 2h 55m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b61023c005c47519e0659b0199cee2a : SUCCESS in 1h 59m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/caf686b884cf494b83adeb6dd1865135 : FAILURE in 15m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/92de00bfb56e4f4c9fa59977c2553111 : SUCCESS in 34m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a353712ed66d48d2bcbbe148968b171b : SUCCESS in 1h 09m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17bee32e99064e9092b7dd5720802a9c : SUCCESS in 1h 43m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/132338417d7646678f42aab981412cc7 : SUCCESS in 40m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e0e03d628c2b4055bde9c265aca4e09a : FAILURE in 1h 54m 22s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a951991e6d154cdb8f450607395f132b : FAILURE in 20m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3bec45a764ec4ca8b20231f1d0407681 : SUCCESS in 1h 52m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2124e0a1033349ba86785412127facee : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":105},{"id":"7e175521dc2ed797cab37d9e2a733a005af94586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 18:33:48.000000000","message":"Patch Set 105:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f2ef6c3cfc49e1b8717c5d83bbf476\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa2294e8ae644b628da7572ae5925b52 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9877a8378234a0894e352ee1dc7f23c : RETRY_LIMIT in 4m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fbd004f4ae3647249ab740828ef91394 : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":105},{"id":"97fbd1b403704ef61539a6bcb4137c4b2945add8","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-26 10:10:32.000000000","message":"Patch Set 105:\n\n(1 comment)","accounts_in_message":[],"_revision_number":105},{"id":"4afe2176b397b0f9f7165d928f6a598516dec9cc","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-03-04 09:23:33.000000000","message":"Patch Set 105:\n\n(1 comment)","accounts_in_message":[],"_revision_number":105},{"id":"4250257158e3903b85fa64ba8453556c05abd181","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-04 09:29:35.000000000","message":"Patch Set 105:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/821f1799a3bd436285b1183521d3b7f8 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/baa9c49659cc4a8697bf5c90dfefc3d1 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":105},{"id":"ee07f239f1cc3a52fc9c515d7efcac05c34b0230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:13:14.000000000","message":"Patch Set 105:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c6b0f395ca4e61ac04dc49eb2884b8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf0738b5aa6e423280b2a3a31b287d86 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63b6c9ae6ae14182b9f5e96ae120f5c5 : RETRY_LIMIT in 5m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b5c0e94f12b4da9bcd062717fe9005e : SUCCESS in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":105},{"id":"047aa52c9c6ff2dbb445b28af418eeb84198446e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 11:53:02.000000000","message":"Patch Set 105:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/189617220b4447718cce3e2052643b2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b477e00ecb84416dbee648858c51b60f : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeb968b031284a1783bd18ea0f90a818 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e17ace562ce24560bfa675cb42916257 : FAILURE in 10m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/939eb64efdbc43f78e46fb88f5f4932e : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e188924f243d425b8dc02449c7df77a5 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/13f0c768e7614128949a0ce3165eb1d6 : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c303f3a0d0e44f88ae81808f901398b9 : SUCCESS in 58m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24b7967bdfb942b6b2ae1cf59d9d2165 : SUCCESS in 1h 21m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5d0c93f3ffb4828a3d47bdf2fd9a6b3 : SUCCESS in 50m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/706e65e8d6754f748158f0712609bc60 : SUCCESS in 17m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e3784042b87444fa775dc9fe9623b4a : SUCCESS in 21m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cfd8f6a71764454b9defbd554f96f1da : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/390261418f0340448c9f339d03a5022a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16822aead4c441ba9c6edb0fe29e99d7 : SUCCESS in 5m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc814995d6be414daf86d8c92fe18c44 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed202ab5681a41a2b5186137b6619dcd : FAILURE in 46m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ca9e6d423814b6597ac4e0993217ba2 : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f30b0ee94a04589be2d7a5659a17503 : SUCCESS in 1h 49m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aabbd9a18cb04613834a145ba34c97f3 : SUCCESS in 1h 37m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5df9c147810448beb7e89c161164e7ca : SUCCESS in 1h 49m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/80fa63375011467091169eca09e07a27 : SUCCESS in 1h 59m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7902449b58da42f3b1e3132ce563e33d : SUCCESS in 38m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a28309c9337844ada11c4b2035c749ee : SUCCESS in 1h 14m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/52c5d337e789475a870876e4c0f4f0b8 : SUCCESS in 1h 29m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f698324b95b04071ae2f170f1662bf63 : SUCCESS in 46m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb343c377d7b4b94ad5c02cf26862461 : SUCCESS in 2h 06m 17s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d64048d69f854516b883415ae544813b : FAILURE in 2h 28m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f30d5a1c07443e5863a218483ec4e3b : SUCCESS in 1h 26m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c8e493ff0c740829469c0df0b029134 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":105},{"id":"677f4e158828bcbcfa9b03df0f759d9e42022961","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-06 15:30:26.000000000","message":"Patch Set 105: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":105},{"id":"505c813c234142f22f5efdef58e98805557ed76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-09 09:20:57.000000000","message":"Uploaded patch set 106: Patch Set 105 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":106},{"id":"28798f440b43e305645307c9a83237427cd7a561","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-09 09:29:00.000000000","message":"Patch Set 106:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4cd6771e100e4dfb9b8a7b04d650499e : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e7577f2ca8344cbeaecb20e06b0f1cfd : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":106},{"id":"9c9190ee58bb43db1f4f221107c12d69a83fe4bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 09:37:58.000000000","message":"Patch Set 106:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f88c95308c49d39c191d34afaf1380\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/616f6235226d4c829ea238457723e6c5 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2fb049158eb47c58971e397886574c2 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/de8fa28fd3524768ac147546d41ca9bd : SUCCESS in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":106},{"id":"93d0ae24e5a3ad51261e9250d309d2b3ec7e47e4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-09 09:47:20.000000000","message":"Patch Set 106:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d430372324644fe1a9b5762e8596ab80 : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":106},{"id":"d785056f9cde529c58fa0af5acfd65810fba11ec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-09 10:06:09.000000000","message":"Patch Set 106:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":106},{"id":"f5d28ef894503230d2f6b226310f9d8547592fa3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-09 10:23:18.000000000","message":"Patch Set 106:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/93309cdeaaa94f72ae0d78376db31aad : FAILURE in 42m 43s","accounts_in_message":[],"_revision_number":106},{"id":"365fe714653b83b76201f1159583d5843574bf3d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-09 10:23:51.000000000","message":"Patch Set 106:\n\nBuild failed.\n\n- hpe-fc-driver build/1f1eb9f19745467d86d4b354b155b864 : FAILURE in 31m 30s","accounts_in_message":[],"_revision_number":106},{"id":"cb55545fd487dcf67b15837a08640f5a8f33cf75","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-09 10:35:28.000000000","message":"Patch Set 106:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":106},{"id":"2bf21b3284658363f58f7936bec3c0c8d6e68350","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-09 10:40:05.000000000","message":"Patch Set 106:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":106},{"id":"6ae8c61630f17a696ee5b5fc6da982ddba02efaa","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-09 10:45:46.000000000","message":"Patch Set 106:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":106},{"id":"e676fea5f12dce671920d649f1c5049d4383085c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 11:46:44.000000000","message":"Patch Set 106: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21f6cebfac2f42959723d6735e89c98a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a323e26aba458bb61e1da2d832e124 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b68ef4cba6eb4d508feee1e12a9da64d : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db4b0357c5d147d1909904b8d74fafdb : SUCCESS in 10m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c6ab8459c1645d9baf6a1dc7b275ca0 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cd8214e09343b7a625651f58f9593f : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c847454c7d5545a5972948189ff4a50c : SUCCESS in 1h 08m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0967d0357ac645e69fc144013e6ea1ca : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81a947bbfcf445b48e0ba4d06f454e58 : SUCCESS in 1h 29m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f129d30c19a4cfa95555e727225d52d : SUCCESS in 1h 06m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea6d512937b84ea891107dfcbbd30bb5 : SUCCESS in 14m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aa3bcd851c484e6db226b3fcfb1a6ae9 : SUCCESS in 24m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7f8878a872a4eb2bc2a8bbe0dd955fa : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34e341c9e21f4a959971141283caa2e4 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4f50ee6e878466a9b4ea13429b404b5 : SUCCESS in 5m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/53d18cdd7be54d1385d3b56802661600 : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa98eaff105d47b1aa845a05ee738c20 : FAILURE in 42m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5509cf111d94e5785f0ec1510a4fee3 : FAILURE in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/372a5566d05e41e78e75427829154a0b : SUCCESS in 1h 54m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d433b9e6d0384149ba30238ec8789bc2 : FAILURE in 2h 17m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/389304f320eb44cca8d395098ec179f1 : SUCCESS in 1h 56m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c065aa9d65e543e39a3371a24306e702 : SUCCESS in 1h 57m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b6d9c9b89c54acab05dc495d6f96334 : SUCCESS in 31m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/302a858b2149449ca083f6493f53feb9 : SUCCESS in 52m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6bee96324d46456c89ec63a8e2195a68 : SUCCESS in 1h 25m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed6748e895f54973bc51d88e1fcf7c09 : SUCCESS in 35m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e7195139248f48e6901a23804db45ce1 : FAILURE in 1h 45m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/74fe97e233dc45c6aea27f1b8420b99a : SUCCESS in 2h 17m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46c0a6b343e1433598d7ee0bf1b6938f : SUCCESS in 1h 13m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d664b9dc25ef4823bc713dd32f5423d1 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":106},{"id":"5d5b0cbf362d09e8585b5d73e9abc43cceb75cca","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-09 12:31:46.000000000","message":"Patch Set 106:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3221/.","accounts_in_message":[],"_revision_number":106},{"id":"00f4c2ca1f2465e761f5e780fe3d78ebbb3ea318","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-09 12:50:17.000000000","message":"Patch Set 106:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3394/.","accounts_in_message":[],"_revision_number":106},{"id":"bf18c23c871c89c9206ee55ec942f1be04b6be4a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-09 14:53:19.000000000","message":"Patch Set 106:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1019/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":106},{"id":"c78214ea14c90773ae99ffc2d6631d0f2c22fe1f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-09 16:20:21.000000000","message":"Patch Set 106:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-106/FC/ : UNSTABLE in 2m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-106/iSCSI/ : FAILURE in 6h 58m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":106},{"id":"0fd50cc0ac611db62b4714b7d4d4933c05b8b458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-12 05:41:09.000000000","message":"Uploaded patch set 107: Patch Set 106 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":107},{"id":"d5543e21fe314519f69adc448b029be7a7975b21","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-12 05:51:30.000000000","message":"Patch Set 107:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9ca7f771354d47efb1ea2959ffac99ba : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/389757903a9a429d86d292b17ff31c48 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":107},{"id":"c8966aa1546421a5c0f309d0bfd126f2e53c34c9","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-12 05:56:04.000000000","message":"Patch Set 107:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3980447cbe994f998631debab9471075 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":107},{"id":"cc3507bcd9637ba9fdaeefbde6275acc7a4ce3e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 06:02:10.000000000","message":"Patch Set 107:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc0f1b68a746468aa29d4a1bd3fec7af\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d638242d13f845e1a5a284bf0b9b278f : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8564d58137be443dbba810ab5c7f592d : FAILURE in 13m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3f5ccf76dc804f02beb3772d748dcba6 : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":107},{"id":"183458ae4ce4765925f1317d0f9172662ca12e61","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-12 06:17:25.000000000","message":"Patch Set 107:\n\nBuild failed.\n\n- hpe-fc-driver build/896d19fbb5de4646af4b274ac7cf7410 : FAILURE in 32m 55s","accounts_in_message":[],"_revision_number":107},{"id":"8b27bad7d5720279a3551f2a88b0924562fd4523","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-12 06:18:46.000000000","message":"Patch Set 107:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7ffe7a7f08e7265250d46bd91f654d2a476b8958 .","accounts_in_message":[],"_revision_number":107},{"id":"8a3bc7d28801424ffb558971c960f1e520fc40ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-12 06:19:45.000000000","message":"Patch Set 107:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 05s","accounts_in_message":[],"_revision_number":107},{"id":"4a60d3a50a34cdfe22df2e44ef63c575dbfe78ba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-12 06:21:26.000000000","message":"Patch Set 107:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/12be70ea5ffb49e2ab08f641bfa45bff : FAILURE in 38m 57s","accounts_in_message":[],"_revision_number":107},{"id":"6e6bd7e6a3aebaf62860db3ce745f6fc804342e9","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-12 06:24:35.000000000","message":"Patch Set 107:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":107},{"id":"c6e1843cf0eb8857fd7207e239084efa70a8fada","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-12 06:49:45.000000000","message":"Patch Set 107:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3277/.","accounts_in_message":[],"_revision_number":107},{"id":"1f4c2f9ec6f5d752e2e8fef8c4c14c05a5ddc1cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-12 06:50:25.000000000","message":"Patch Set 107:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3450/.","accounts_in_message":[],"_revision_number":107},{"id":"f724b6ab8a1dfc0a0e7ca05e091bd65994dc34a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 08:22:31.000000000","message":"Patch Set 107: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925454a2d5e34601b3b66e0453899cd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83152e1a6f8748d8a5ac5222ece9de7f : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3847dc4005d24c189aff73085210e114 : SUCCESS in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d188d4c1ee644a44b0618a6bd9ec4796 : SUCCESS in 7m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70b39f92b5c74c53a7b7c5c74d10df66 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f44a1e80af403fa895f8238dff3210 : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2b8e24cfafa74b19921e1648541a9f02 : SUCCESS in 44m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62cfff33a1ff45169861131edd034983 : SUCCESS in 1h 20m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b017befcc9224ae28d28bb00aecdf708 : SUCCESS in 1h 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b952be3cb30b40a8a1e68ab59261bc64 : SUCCESS in 1h 01m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b8e3533192f4aa0a3eb911cfccbb29a : SUCCESS in 12m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4eee54490cf4cf2a5081e7a088655da : SUCCESS in 22m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adb4296ae61f44fda18cc6767abca098 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2cfb77c781b443b4a61bf99341bbe1d5 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfd8f3a1b1ba451288de950b14c5d183 : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2c8a04ba3c246ae9c3d4a2ffa055d1a : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8615e2d203fe46b7a9ee6a9b3e9dde9b : FAILURE in 41m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8d528b0375d4d02a21ef3a3c6cc37d3 : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/863f575836604cf1adf9641d48da2e5c : SUCCESS in 1h 36m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/19226d8c38694e85bfbb26e03deb0ee6 : SUCCESS in 2h 33m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a0c103dd07a44ebbfac264af5c85cd0 : SUCCESS in 1h 59m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9bfcdacf8d814298a13a915a9e1ac895 : SUCCESS in 1h 44m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/63f9bef47acd4daca84c9bf67d11930a : SUCCESS in 36m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7962e0c2de06420688f28fbfe9aa6cfd : SUCCESS in 39m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de41e06fc31f4dfb9dd9f10f6d28aea0 : SUCCESS in 1h 27m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32ea0173d0354db8827dae4df92c08d2 : SUCCESS in 45m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7d0434b81c944bd8b500c08a3207e7f3 : SUCCESS in 1h 41m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/33040e395f9c40a396076be69f0e25e9 : POST_FAILURE in 1h 44m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c527f90e13e441f58e291fef8a97659d : SUCCESS in 50m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/235e7d9471994922897eadcc2b9b8dc4 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":107},{"id":"3f71c0b7d7cc4f946d9140d2203bc76795b7fd5b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-12 20:06:35.000000000","message":"Patch Set 107:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1079/index.html : SUCCESS in 95m","accounts_in_message":[],"_revision_number":107},{"id":"30aeca1c6b0adde010cbad37ab0d2d448f121960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-17 10:55:03.000000000","message":"Uploaded patch set 108: Patch Set 107 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":108},{"id":"fad00145407a96174b6f339ffa9766535b0f5eaf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 11:10:02.000000000","message":"Patch Set 108:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ab11dbb642944984bf73256ee3f50fb3 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d1e9b5b349844681a70592778e255843 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":108},{"id":"f98dfd2f172e398068b492488f56c5030a75b938","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-17 11:12:21.000000000","message":"Patch Set 108:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/bbe390b00f7647369b4e3b9ecbf1cf48 : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":108},{"id":"0d8837fa8ae0f8b14542796e074394260275a9be","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-17 11:32:48.000000000","message":"Patch Set 108:\n\nBuild failed.\n\n- hpe-fc-driver build/cf14bbcc8a0a499483e459b5009025f2 : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":108},{"id":"f6e15a3766a9c5c78ca9bfe431771c8d29c932f6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-17 12:08:17.000000000","message":"Patch Set 108:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/938771 : FAILURE in 37m 50s","accounts_in_message":[],"_revision_number":108},{"id":"a9b9a055ef03f8d086c5fffd10950f1d493a5fb0","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-17 12:21:09.000000000","message":"Patch Set 108:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/938771 : SUCCESS","accounts_in_message":[],"_revision_number":108},{"id":"4462e07e7a296b9bda5206b1b2fbf399178360f3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-17 14:01:51.000000000","message":"Patch Set 108:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-108/FC/ : UNSTABLE in 13m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-938771-108/iSCSI/ : FAILURE in 3h 6m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":108},{"id":"23809b00d27608507870884e0cdd33055c8b435b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 17:23:14.000000000","message":"Patch Set 108:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3510/.","accounts_in_message":[],"_revision_number":108},{"id":"053fb7556fa090872e747fac7a2fde2c0eeac746","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 18:43:28.000000000","message":"Patch Set 108:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3337/.","accounts_in_message":[],"_revision_number":108},{"id":"3834cf75b050a4ee64ee1f6bcbebe995eaa8af50","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-04-09 11:15:05.000000000","message":"Abandoned\n\nCreated a new patch as this patch reached its limit.","accounts_in_message":[],"_revision_number":108}],"current_revision_number":108,"current_revision":"3a8dc22373dcaf059a21321b7084cfce762c08ca","revisions":{"60883bbc3bd700ce341a46b0dca4381deaff2c22":{"kind":"REWORK","_number":1,"created":"2025-01-09 07:08:19.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/1"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60883bbc3bd700ce341a46b0dca4381deaff2c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60883bbc3bd700ce341a46b0dca4381deaff2c22"}]},"branch":"refs/heads/master"},"0569ecc126ea1fc56f137093a8b18b227e9168cd":{"kind":"REWORK","_number":2,"created":"2025-01-09 07:10:02.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/2"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:10:02.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0569ecc126ea1fc56f137093a8b18b227e9168cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0569ecc126ea1fc56f137093a8b18b227e9168cd"}]},"branch":"refs/heads/master"},"31bf707dbad24f7334a1048c31860c74c90faf4e":{"kind":"REWORK","_number":3,"created":"2025-01-09 07:12:40.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/3"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:12:39.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31bf707dbad24f7334a1048c31860c74c90faf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31bf707dbad24f7334a1048c31860c74c90faf4e"}]},"branch":"refs/heads/master"},"5f7dd4bd35b786a6c3da03f08989d459c5a4cd79":{"kind":"REWORK","_number":4,"created":"2025-01-09 08:53:07.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/4"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 08:53:06.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f7dd4bd35b786a6c3da03f08989d459c5a4cd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f7dd4bd35b786a6c3da03f08989d459c5a4cd79"}]},"branch":"refs/heads/master"},"0dfaed50aab411374c74b09e6f059b63f3e6c6e9":{"kind":"REWORK","_number":5,"created":"2025-01-09 09:03:40.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/5"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 09:03:39.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dfaed50aab411374c74b09e6f059b63f3e6c6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dfaed50aab411374c74b09e6f059b63f3e6c6e9"}]},"branch":"refs/heads/master"},"dff2c67771b778e12752b590e1c846bc063e70cf":{"kind":"REWORK","_number":6,"created":"2025-01-09 09:08:02.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/6"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 09:08:02.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dff2c67771b778e12752b590e1c846bc063e70cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dff2c67771b778e12752b590e1c846bc063e70cf"}]},"branch":"refs/heads/master"},"1926239780a6613faf77d4198df176de04329d4a":{"kind":"REWORK","_number":7,"created":"2025-01-09 09:13:52.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/7"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 09:13:51.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1926239780a6613faf77d4198df176de04329d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1926239780a6613faf77d4198df176de04329d4a"}]},"branch":"refs/heads/master"},"e2ddf7f4503f5683b3453026437a3050ff2dc12c":{"kind":"REWORK","_number":8,"created":"2025-01-14 05:33:18.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/8"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-14 05:33:18.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2ddf7f4503f5683b3453026437a3050ff2dc12c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2ddf7f4503f5683b3453026437a3050ff2dc12c"}]},"branch":"refs/heads/master"},"4443934b0e9b393629edac237768768edcf47456":{"kind":"REWORK","_number":9,"created":"2025-01-14 09:16:09.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/9"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-14 09:16:09.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4443934b0e9b393629edac237768768edcf47456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4443934b0e9b393629edac237768768edcf47456"}]},"branch":"refs/heads/master"},"7f7d2b232302185ed97081bc46e1c90b7973ab1a":{"kind":"REWORK","_number":10,"created":"2025-01-14 11:46:31.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/10"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-14 11:46:31.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f7d2b232302185ed97081bc46e1c90b7973ab1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f7d2b232302185ed97081bc46e1c90b7973ab1a"}]},"branch":"refs/heads/master"},"a8f1485016acb3eacd223fffd5a6bd61be49657c":{"kind":"REWORK","_number":11,"created":"2025-01-14 11:48:33.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/11"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-14 11:48:33.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8f1485016acb3eacd223fffd5a6bd61be49657c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8f1485016acb3eacd223fffd5a6bd61be49657c"}]},"branch":"refs/heads/master"},"c931101c348e0ad0e49d13a77b9920d5ef2317d3":{"kind":"REWORK","_number":12,"created":"2025-01-15 07:00:26.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/12"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-15 07:00:25.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c931101c348e0ad0e49d13a77b9920d5ef2317d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c931101c348e0ad0e49d13a77b9920d5ef2317d3"}]},"branch":"refs/heads/master"},"8cb1774101f7aa32be5e721ce9093163c04438b2":{"kind":"REWORK","_number":13,"created":"2025-01-20 05:48:25.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/13"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-20 05:47:11.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb1774101f7aa32be5e721ce9093163c04438b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb1774101f7aa32be5e721ce9093163c04438b2"}]},"branch":"refs/heads/master"},"1d9c90249706596bad0afe8f5da033aec6eda9f8":{"kind":"REWORK","_number":14,"created":"2025-01-20 05:51:24.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/14"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-20 05:51:24.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9c90249706596bad0afe8f5da033aec6eda9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9c90249706596bad0afe8f5da033aec6eda9f8"}]},"branch":"refs/heads/master"},"5853742aeebce5533c55ad10007033eb2953ba8e":{"kind":"REWORK","_number":15,"created":"2025-01-27 09:47:59.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/15"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-27 09:47:50.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5853742aeebce5533c55ad10007033eb2953ba8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5853742aeebce5533c55ad10007033eb2953ba8e"}]},"branch":"refs/heads/master"},"4561f004254b7f1da2f7e30bc5dc5b2802dff26f":{"kind":"REWORK","_number":16,"created":"2025-01-27 11:28:54.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/16"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-27 11:28:53.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4561f004254b7f1da2f7e30bc5dc5b2802dff26f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4561f004254b7f1da2f7e30bc5dc5b2802dff26f"}]},"branch":"refs/heads/master"},"4724640474cb7b710b6c77523739f7af640b1554":{"kind":"REWORK","_number":17,"created":"2025-01-29 06:43:38.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/17"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-29 06:43:38.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4724640474cb7b710b6c77523739f7af640b1554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4724640474cb7b710b6c77523739f7af640b1554"}]},"branch":"refs/heads/master"},"430ac80854b0b24a8273364a316ddc897b48f785":{"kind":"REWORK","_number":18,"created":"2025-01-29 10:51:40.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/18"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-29 10:51:40.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/430ac80854b0b24a8273364a316ddc897b48f785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/430ac80854b0b24a8273364a316ddc897b48f785"}]},"branch":"refs/heads/master"},"70fdd5e269a303d58c8822ebfb8c5f8c1727a25b":{"kind":"REWORK","_number":19,"created":"2025-01-29 10:58:00.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/19"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-01-29 10:58:00.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70fdd5e269a303d58c8822ebfb8c5f8c1727a25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70fdd5e269a303d58c8822ebfb8c5f8c1727a25b"}]},"branch":"refs/heads/master"},"d1079cbce3fd6e7fe7e5f5baabf96bde7d02dd68":{"kind":"REWORK","_number":20,"created":"2025-05-09 10:32:31.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/20"}}},"commit":{"parents":[{"commit":"83d3834eee73420429e9d5622e629f0d0e5fcab6","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83d3834eee73420429e9d5622e629f0d0e5fcab6"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-05-09 15:53:54.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1079cbce3fd6e7fe7e5f5baabf96bde7d02dd68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1079cbce3fd6e7fe7e5f5baabf96bde7d02dd68"}]},"branch":"refs/heads/master"},"bd4cff3b1f6709c61d287acee83d8b28857370eb":{"kind":"REWORK","_number":21,"created":"2025-05-20 10:22:30.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/21"}}},"commit":{"parents":[{"commit":"d23a17da679e1180a546d78138e257f397fe4b4b","subject":"Merge \"nfs: minimal mypy coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23a17da679e1180a546d78138e257f397fe4b4b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"dakshina","email":"Dakshina.Purohit@ibm.com","date":"2025-05-20 15:52:31.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd4cff3b1f6709c61d287acee83d8b28857370eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd4cff3b1f6709c61d287acee83d8b28857370eb"}]},"branch":"refs/heads/master"},"ae28abfbc174c196f4141b7dabd331487dcd5020":{"kind":"REWORK","_number":22,"created":"2025-06-02 11:06:54.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/22"}}},"commit":{"parents":[{"commit":"26da7ba95eba679c8f80f4be90c20debf9c79ed1","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26da7ba95eba679c8f80f4be90c20debf9c79ed1"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-02 10:52:30.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae28abfbc174c196f4141b7dabd331487dcd5020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae28abfbc174c196f4141b7dabd331487dcd5020"}]},"branch":"refs/heads/master"},"a6d5feaa7d9fcddd035c023c99f96510daa1fea8":{"kind":"REWORK","_number":23,"created":"2025-06-02 11:45:40.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/23"}}},"commit":{"parents":[{"commit":"534202e357c008cd1722b642fcecbc9a1b85b1da","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/534202e357c008cd1722b642fcecbc9a1b85b1da"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-02 11:45:40.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6d5feaa7d9fcddd035c023c99f96510daa1fea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6d5feaa7d9fcddd035c023c99f96510daa1fea8"}]},"branch":"refs/heads/master","description":"Rebase"},"5c9432f1e3fc33528a5402646eb39f16e3dad452":{"kind":"REWORK","_number":24,"created":"2025-06-02 11:48:47.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/24"}}},"commit":{"parents":[{"commit":"534202e357c008cd1722b642fcecbc9a1b85b1da","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/534202e357c008cd1722b642fcecbc9a1b85b1da"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-02 11:48:46.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c9432f1e3fc33528a5402646eb39f16e3dad452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c9432f1e3fc33528a5402646eb39f16e3dad452"}]},"branch":"refs/heads/master"},"9ae7ff6fc3d602095723c277f5a176673603042a":{"kind":"REWORK","_number":25,"created":"2025-06-03 06:29:50.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/25"}}},"commit":{"parents":[{"commit":"534202e357c008cd1722b642fcecbc9a1b85b1da","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/534202e357c008cd1722b642fcecbc9a1b85b1da"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 06:29:50.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ae7ff6fc3d602095723c277f5a176673603042a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ae7ff6fc3d602095723c277f5a176673603042a"}]},"branch":"refs/heads/master"},"abb2cd93bb10b845d27f3471818327ecf1676c63":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-06-03 06:38:28.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/26"}}},"commit":{"parents":[{"commit":"8aadbde81900e04b0376dba0e36553b9161e0ade","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8aadbde81900e04b0376dba0e36553b9161e0ade"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 06:38:28.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/abb2cd93bb10b845d27f3471818327ecf1676c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/abb2cd93bb10b845d27f3471818327ecf1676c63"}]},"branch":"refs/heads/master","description":"Rebase"},"03ff5cb5c4547521f413a491443f652a6a97d2ea":{"kind":"REWORK","_number":27,"created":"2025-06-03 10:18:03.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/27"}}},"commit":{"parents":[{"commit":"48eae0418c93a575e4505d3b79eb2bdb7b3ba5e0","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48eae0418c93a575e4505d3b79eb2bdb7b3ba5e0"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 10:18:03.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03ff5cb5c4547521f413a491443f652a6a97d2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03ff5cb5c4547521f413a491443f652a6a97d2ea"}]},"branch":"refs/heads/master","description":"Rebase"},"96849c0c47d1c2b6ebe71e2c7154be8fa8e3b219":{"kind":"REWORK","_number":28,"created":"2025-06-03 10:19:25.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/28"}}},"commit":{"parents":[{"commit":"48eae0418c93a575e4505d3b79eb2bdb7b3ba5e0","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48eae0418c93a575e4505d3b79eb2bdb7b3ba5e0"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 10:19:25.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96849c0c47d1c2b6ebe71e2c7154be8fa8e3b219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96849c0c47d1c2b6ebe71e2c7154be8fa8e3b219"}]},"branch":"refs/heads/master"},"af0c083865dbb89282f559db8a396a7072685846":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2025-06-03 16:54:13.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/29"}}},"commit":{"parents":[{"commit":"1af948ee53ee5d0df878769d970956442fac6481","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1af948ee53ee5d0df878769d970956442fac6481"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 16:54:13.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af0c083865dbb89282f559db8a396a7072685846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af0c083865dbb89282f559db8a396a7072685846"}]},"branch":"refs/heads/master","description":"Rebase"},"ae1dfb2ee1c45552b91b41ccfab47f4355ea6425":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-06-03 17:18:11.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/30"}}},"commit":{"parents":[{"commit":"3bf3a41f0ffc50da4f1e041428026977c33a1699","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bf3a41f0ffc50da4f1e041428026977c33a1699"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-03 17:18:11.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1dfb2ee1c45552b91b41ccfab47f4355ea6425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1dfb2ee1c45552b91b41ccfab47f4355ea6425"}]},"branch":"refs/heads/master","description":"Rebase"},"5581ddc805c055d7bfd8a53069f18a5053b1423e":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2025-06-04 06:18:30.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/31"}}},"commit":{"parents":[{"commit":"134025576e3772cae6534f3d482d4ab97123d6e7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134025576e3772cae6534f3d482d4ab97123d6e7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-04 06:18:30.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5581ddc805c055d7bfd8a53069f18a5053b1423e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5581ddc805c055d7bfd8a53069f18a5053b1423e"}]},"branch":"refs/heads/master","description":"Rebase"},"792cd4596e3cfdae1264f984b6e85ca8eda8e805":{"kind":"REWORK","_number":32,"created":"2025-06-04 07:56:06.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/32"}}},"commit":{"parents":[{"commit":"134025576e3772cae6534f3d482d4ab97123d6e7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134025576e3772cae6534f3d482d4ab97123d6e7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-04 07:56:06.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/792cd4596e3cfdae1264f984b6e85ca8eda8e805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/792cd4596e3cfdae1264f984b6e85ca8eda8e805"}]},"branch":"refs/heads/master"},"e111c227ab493621216087d193a984f5a87c9f44":{"kind":"REWORK","_number":33,"created":"2025-06-04 17:18:09.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/33"}}},"commit":{"parents":[{"commit":"134025576e3772cae6534f3d482d4ab97123d6e7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134025576e3772cae6534f3d482d4ab97123d6e7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-04 17:18:09.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e111c227ab493621216087d193a984f5a87c9f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e111c227ab493621216087d193a984f5a87c9f44"}]},"branch":"refs/heads/master"},"3228c65e0d0636af252712ae301187b6326991e9":{"kind":"REWORK","_number":34,"created":"2025-06-05 06:42:48.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/34"}}},"commit":{"parents":[{"commit":"134025576e3772cae6534f3d482d4ab97123d6e7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134025576e3772cae6534f3d482d4ab97123d6e7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 06:42:48.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3228c65e0d0636af252712ae301187b6326991e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3228c65e0d0636af252712ae301187b6326991e9"}]},"branch":"refs/heads/master"},"dcc40441c30157832380b8ff6e22dc9e92a44955":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2025-06-05 10:52:29.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/35"}}},"commit":{"parents":[{"commit":"343654ffc4cec2012600649c0e8a0ff6e4aa7454","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/343654ffc4cec2012600649c0e8a0ff6e4aa7454"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-06-05 10:52:29.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcc40441c30157832380b8ff6e22dc9e92a44955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcc40441c30157832380b8ff6e22dc9e92a44955"}]},"branch":"refs/heads/master","description":"Rebase"},"2e60fb53e685628b630b0ae0d4550ebd487d4b46":{"kind":"REWORK","_number":36,"created":"2025-06-05 11:52:51.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/36"}}},"commit":{"parents":[{"commit":"c48d0c9db6ee0fed29f7f5b9ae180d20d49ca301","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c48d0c9db6ee0fed29f7f5b9ae180d20d49ca301"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-06-05 11:49:47.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e60fb53e685628b630b0ae0d4550ebd487d4b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e60fb53e685628b630b0ae0d4550ebd487d4b46"}]},"branch":"refs/heads/master"},"ce13494620ac7c89a01bc27694a3064e4fca0c6a":{"kind":"NO_CHANGE","_number":37,"created":"2025-06-05 12:02:56.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/37"}}},"commit":{"parents":[{"commit":"22ee358c50b9d27d0cb5adc7043bda3c2cf6747b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22ee358c50b9d27d0cb5adc7043bda3c2cf6747b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 12:00:59.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce13494620ac7c89a01bc27694a3064e4fca0c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce13494620ac7c89a01bc27694a3064e4fca0c6a"}]},"branch":"refs/heads/master"},"1e61aa709e95fe43d1f5189dbc02912f50f2450d":{"kind":"REWORK","_number":38,"created":"2025-06-05 19:11:20.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/38"}}},"commit":{"parents":[{"commit":"22ee358c50b9d27d0cb5adc7043bda3c2cf6747b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22ee358c50b9d27d0cb5adc7043bda3c2cf6747b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 19:10:51.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e61aa709e95fe43d1f5189dbc02912f50f2450d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e61aa709e95fe43d1f5189dbc02912f50f2450d"}]},"branch":"refs/heads/master"},"5f5c1834af720461cd1c5da2902c10ffaef2eb7a":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2025-06-05 19:13:45.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/39"}}},"commit":{"parents":[{"commit":"050a937a9ab867da8d9341c30cfef34a91afbcbb","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/050a937a9ab867da8d9341c30cfef34a91afbcbb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 19:13:45.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f5c1834af720461cd1c5da2902c10ffaef2eb7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f5c1834af720461cd1c5da2902c10ffaef2eb7a"}]},"branch":"refs/heads/master","description":"Rebase"},"1ab9cd5fdf60de83d25c3a41f44f1d6651f40f21":{"kind":"NO_CHANGE","_number":40,"created":"2025-06-06 11:40:30.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/40"}}},"commit":{"parents":[{"commit":"60058dd63067ee9a9a915b256d8098132b6e6335","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60058dd63067ee9a9a915b256d8098132b6e6335"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-06-06 10:47:55.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ab9cd5fdf60de83d25c3a41f44f1d6651f40f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ab9cd5fdf60de83d25c3a41f44f1d6651f40f21"}]},"branch":"refs/heads/master"},"eb1bd2ee643d0aec0accbda371d5ca6fb42a52b0":{"kind":"NO_CHANGE","_number":41,"created":"2025-06-09 10:53:50.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/41"}}},"commit":{"parents":[{"commit":"7bb69a6f8b6ebeda044cf3a13e3c066dacf8d47a","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bb69a6f8b6ebeda044cf3a13e3c066dacf8d47a"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-01-09 07:05:33.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-06-09 10:39:53.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb1bd2ee643d0aec0accbda371d5ca6fb42a52b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb1bd2ee643d0aec0accbda371d5ca6fb42a52b0"}]},"branch":"refs/heads/master"},"ca74a3b764de7dc6a138f6e961afaa2f632d2d68":{"kind":"NO_CHANGE","_number":42,"created":"2025-06-09 14:35:39.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/42"}}},"commit":{"parents":[{"commit":"3da852f3f4dbd3a43fcb22ea641b1028ea4bb279","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3da852f3f4dbd3a43fcb22ea641b1028ea4bb279"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca74a3b764de7dc6a138f6e961afaa2f632d2d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca74a3b764de7dc6a138f6e961afaa2f632d2d68"}]},"branch":"refs/heads/master"},"200464f4ab8c61209279bac2e49dba70646d708c":{"kind":"REWORK","_number":43,"created":"2025-06-09 15:05:24.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/43"}}},"commit":{"parents":[{"commit":"3da852f3f4dbd3a43fcb22ea641b1028ea4bb279","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3da852f3f4dbd3a43fcb22ea641b1028ea4bb279"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 15:04:46.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/200464f4ab8c61209279bac2e49dba70646d708c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/200464f4ab8c61209279bac2e49dba70646d708c"}]},"branch":"refs/heads/master"},"26b5b0eb0709c58565c3d8d67b9778286b6fabbd":{"kind":"REWORK","_number":44,"created":"2025-06-10 06:46:09.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/44"}}},"commit":{"parents":[{"commit":"3da852f3f4dbd3a43fcb22ea641b1028ea4bb279","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3da852f3f4dbd3a43fcb22ea641b1028ea4bb279"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-10 06:46:09.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26b5b0eb0709c58565c3d8d67b9778286b6fabbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26b5b0eb0709c58565c3d8d67b9778286b6fabbd"}]},"branch":"refs/heads/master"},"a998c151eda891bc4cb330c2cc47106ef234bc35":{"kind":"REWORK","_number":45,"created":"2025-06-10 09:59:24.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/45"}}},"commit":{"parents":[{"commit":"3da852f3f4dbd3a43fcb22ea641b1028ea4bb279","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3da852f3f4dbd3a43fcb22ea641b1028ea4bb279"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-10 09:59:24.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a998c151eda891bc4cb330c2cc47106ef234bc35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a998c151eda891bc4cb330c2cc47106ef234bc35"}]},"branch":"refs/heads/master"},"d67f98b105a11dff41c60d3e6fa4f8179c364fce":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2025-06-10 16:11:05.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/46"}}},"commit":{"parents":[{"commit":"0242128369768d65567129f48e58f9f990232b8e","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0242128369768d65567129f48e58f9f990232b8e"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-10 16:11:05.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67f98b105a11dff41c60d3e6fa4f8179c364fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67f98b105a11dff41c60d3e6fa4f8179c364fce"}]},"branch":"refs/heads/master","description":"Rebase"},"6b5182be6c44e8dffba0ce6b3a89dba56f7f4432":{"kind":"NO_CHANGE","_number":47,"created":"2025-06-12 08:53:15.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/47"}}},"commit":{"parents":[{"commit":"0250e72c16031520500a984a6e8b00af40d44181","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0250e72c16031520500a984a6e8b00af40d44181"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-11 10:16:23.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b5182be6c44e8dffba0ce6b3a89dba56f7f4432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b5182be6c44e8dffba0ce6b3a89dba56f7f4432"}]},"branch":"refs/heads/master"},"2f9fe6a2dd82e8599b4792a506d58525e9c306c6":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2025-06-20 09:37:10.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/48"}}},"commit":{"parents":[{"commit":"60058dd63067ee9a9a915b256d8098132b6e6335","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60058dd63067ee9a9a915b256d8098132b6e6335"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-06-10 16:08:32.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f9fe6a2dd82e8599b4792a506d58525e9c306c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f9fe6a2dd82e8599b4792a506d58525e9c306c6"}]},"branch":"refs/heads/master"},"8ba3e59a4e3c2b1c762c2a283acf977655cc3a18":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2025-06-20 09:39:54.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/49"}}},"commit":{"parents":[{"commit":"03e9e7f91491d89ada547fe31d0fa84561dc03ac","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e9e7f91491d89ada547fe31d0fa84561dc03ac"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Nimesh Desai","email":"Nimesh.Desai@ibm.com","date":"2025-06-20 09:39:54.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ba3e59a4e3c2b1c762c2a283acf977655cc3a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ba3e59a4e3c2b1c762c2a283acf977655cc3a18"}]},"branch":"refs/heads/master","description":"Rebase"},"ed44d9298284d6d4cfbd0e46a2b338197507d4c1":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2025-06-25 09:49:34.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/50"}}},"commit":{"parents":[{"commit":"7e1ac1176c3f8ff891ee61ef5fd2e3f64cb22761","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e1ac1176c3f8ff891ee61ef5fd2e3f64cb22761"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Nimesh Desai","email":"Nimesh.Desai@ibm.com","date":"2025-06-25 09:49:34.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed44d9298284d6d4cfbd0e46a2b338197507d4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed44d9298284d6d4cfbd0e46a2b338197507d4c1"}]},"branch":"refs/heads/master","description":"Rebase"},"43d0c5aad811cbe9995d779a588f8802adddb943":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2025-07-01 05:28:15.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/51"}}},"commit":{"parents":[{"commit":"3bdc54ec8396a561583ff64bad8c2e2e2cfd4dbd","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bdc54ec8396a561583ff64bad8c2e2e2cfd4dbd"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Nimesh Desai","email":"Nimesh.Desai@ibm.com","date":"2025-07-01 05:28:15.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43d0c5aad811cbe9995d779a588f8802adddb943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43d0c5aad811cbe9995d779a588f8802adddb943"}]},"branch":"refs/heads/master","description":"Rebase"},"7838016965d80690f6a24afb26060f63c2b8bfb3":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2025-07-01 08:23:30.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/52"}}},"commit":{"parents":[{"commit":"9688d1ceb9f96f5200d01cfc2cbadd99fa148c74","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9688d1ceb9f96f5200d01cfc2cbadd99fa148c74"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-06-30 07:21:34.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7838016965d80690f6a24afb26060f63c2b8bfb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7838016965d80690f6a24afb26060f63c2b8bfb3"}]},"branch":"refs/heads/master"},"b90528bc942c9cd3a5edd211c4e8d9ef6cf29ee4":{"kind":"NO_CODE_CHANGE","_number":53,"created":"2025-07-08 07:15:23.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/53"}}},"commit":{"parents":[{"commit":"9688d1ceb9f96f5200d01cfc2cbadd99fa148c74","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9688d1ceb9f96f5200d01cfc2cbadd99fa148c74"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-07-08 07:15:23.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b90528bc942c9cd3a5edd211c4e8d9ef6cf29ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b90528bc942c9cd3a5edd211c4e8d9ef6cf29ee4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"31c6ac7bebff2a87532c81bac3d0e11541731922":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2025-07-08 07:15:30.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/54"}}},"commit":{"parents":[{"commit":"dc79f1d3ed24c211a94a7cacfc4adc3106e1f41e","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc79f1d3ed24c211a94a7cacfc4adc3106e1f41e"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-07-08 07:15:30.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31c6ac7bebff2a87532c81bac3d0e11541731922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31c6ac7bebff2a87532c81bac3d0e11541731922"}]},"branch":"refs/heads/master","description":"Rebase"},"473b5f0537b6c6d2f01a9486e51b4f6f2b26c9eb":{"kind":"NO_CHANGE","_number":55,"created":"2025-07-08 08:26:53.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/55"}}},"commit":{"parents":[{"commit":"8b56bd21f0df9b1b318de39abce6966e43a0e163","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b56bd21f0df9b1b318de39abce6966e43a0e163"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-08 08:19:39.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/473b5f0537b6c6d2f01a9486e51b4f6f2b26c9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/473b5f0537b6c6d2f01a9486e51b4f6f2b26c9eb"}]},"branch":"refs/heads/master"},"d3dbfe0e10e129a8b89a07aaeb9656340c5960ab":{"kind":"NO_CHANGE","_number":56,"created":"2025-07-08 08:44:42.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/56"}}},"commit":{"parents":[{"commit":"e0af6622e8117449232c7e761c8facb752332f4a","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0af6622e8117449232c7e761c8facb752332f4a"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-07-08 08:42:56.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3dbfe0e10e129a8b89a07aaeb9656340c5960ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3dbfe0e10e129a8b89a07aaeb9656340c5960ab"}]},"branch":"refs/heads/master"},"4b4b8e4fad0b0591973da02b9a8ed114e1734c83":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2025-07-17 10:09:32.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/57"}}},"commit":{"parents":[{"commit":"ae83edf4a449d01dc42f51bbc2bd5ea835de13e4","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae83edf4a449d01dc42f51bbc2bd5ea835de13e4"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Nimesh Desai","email":"Nimesh.Desai@ibm.com","date":"2025-07-17 10:09:32.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b4b8e4fad0b0591973da02b9a8ed114e1734c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b4b8e4fad0b0591973da02b9a8ed114e1734c83"}]},"branch":"refs/heads/master","description":"Rebase"},"b06af18b19d0ceeede238e4898d9d5468433bb6f":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2025-07-18 07:25:16.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/58"}}},"commit":{"parents":[{"commit":"b04411967755f5c2b239e92be747813023be6962","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b04411967755f5c2b239e92be747813023be6962"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-18 07:23:36.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b06af18b19d0ceeede238e4898d9d5468433bb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b06af18b19d0ceeede238e4898d9d5468433bb6f"}]},"branch":"refs/heads/master"},"c6caee7daabfe331a683df2a7c503307c830d1b4":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2025-07-18 08:33:19.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/59"}}},"commit":{"parents":[{"commit":"a3416653f800421b0b0dc64c24293cd37ebca57c","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3416653f800421b0b0dc64c24293cd37ebca57c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-07-18 08:30:34.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6caee7daabfe331a683df2a7c503307c830d1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6caee7daabfe331a683df2a7c503307c830d1b4"}]},"branch":"refs/heads/master"},"6d4bcc61312e0b30bc43d1f50696efd75fc1bd4f":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2025-07-18 13:05:41.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/60"}}},"commit":{"parents":[{"commit":"4b40ab163e1a2a2c4fa9ded8ab4edef8076b7bdd","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b40ab163e1a2a2c4fa9ded8ab4edef8076b7bdd"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-07-18 12:49:21.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d4bcc61312e0b30bc43d1f50696efd75fc1bd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d4bcc61312e0b30bc43d1f50696efd75fc1bd4f"}]},"branch":"refs/heads/master"},"6591ab9c9461b37e7b41831a0fcafcab68493e7a":{"kind":"NO_CHANGE","_number":61,"created":"2025-07-18 14:59:15.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/61"}}},"commit":{"parents":[{"commit":"355e3977fb6b266c00a8ab25b4ef0468e5897aeb","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/355e3977fb6b266c00a8ab25b4ef0468e5897aeb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-07-18 14:44:55.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6591ab9c9461b37e7b41831a0fcafcab68493e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6591ab9c9461b37e7b41831a0fcafcab68493e7a"}]},"branch":"refs/heads/master"},"39532c1d8815c098a80afe1e97666fef25e18dac":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2025-07-21 12:03:14.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/62"}}},"commit":{"parents":[{"commit":"798b2ef07be7f9ac5465d49e9c4014201d56f2c6","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/798b2ef07be7f9ac5465d49e9c4014201d56f2c6"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-07-21 11:13:54.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39532c1d8815c098a80afe1e97666fef25e18dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39532c1d8815c098a80afe1e97666fef25e18dac"}]},"branch":"refs/heads/master"},"f0e77040f68519c03fb3e9bfcdcfd5d3ed344359":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2025-08-05 07:11:59.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/63"}}},"commit":{"parents":[{"commit":"1053f07b5a3ec145f9a30604440aac287d6ae79a","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1053f07b5a3ec145f9a30604440aac287d6ae79a"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-05 07:11:59.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0e77040f68519c03fb3e9bfcdcfd5d3ed344359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0e77040f68519c03fb3e9bfcdcfd5d3ed344359"}]},"branch":"refs/heads/master","description":"Rebase"},"08c98351ae87ea1d75fc547e6a5aecd1c41352e0":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2025-08-07 02:05:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/64"}}},"commit":{"parents":[{"commit":"2f8987f90a84c8d34a26f9699beab42a8a1fd38b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f8987f90a84c8d34a26f9699beab42a8a1fd38b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-07 02:05:51.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08c98351ae87ea1d75fc547e6a5aecd1c41352e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08c98351ae87ea1d75fc547e6a5aecd1c41352e0"}]},"branch":"refs/heads/master","description":"Rebase"},"0496998a730a0d3509ab30b675bdcdb4a2b6a7a1":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2025-08-21 03:32:09.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/65"}}},"commit":{"parents":[{"commit":"ee6a9030025f7ae7babe77473b3531943a4841a0","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee6a9030025f7ae7babe77473b3531943a4841a0"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-30 06:58:47.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0496998a730a0d3509ab30b675bdcdb4a2b6a7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0496998a730a0d3509ab30b675bdcdb4a2b6a7a1"}]},"branch":"refs/heads/master"},"12d71d546d6ed2638794f7e5b3029051ec024939":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2025-08-21 09:10:54.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/66"}}},"commit":{"parents":[{"commit":"7e871397785897db3c79445dd5e9ab7546c30a96","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e871397785897db3c79445dd5e9ab7546c30a96"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-08-14 12:03:06.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12d71d546d6ed2638794f7e5b3029051ec024939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12d71d546d6ed2638794f7e5b3029051ec024939"}]},"branch":"refs/heads/master"},"beb4fbcc45573119e7dc59de0f302474aaad5b5b":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2025-08-25 18:10:54.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/67"}}},"commit":{"parents":[{"commit":"78f04a20a26e1b2be731c38deb4dcad7d15b26a3","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78f04a20a26e1b2be731c38deb4dcad7d15b26a3"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:49:53.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/beb4fbcc45573119e7dc59de0f302474aaad5b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/beb4fbcc45573119e7dc59de0f302474aaad5b5b"}]},"branch":"refs/heads/master"},"cac45646b2c8929b3f5702d9d6269f85c4f0193a":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2025-09-03 05:57:21.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/68"}}},"commit":{"parents":[{"commit":"59a4c99cae5fd3304aae87671b8ce7ff5ff94058","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59a4c99cae5fd3304aae87671b8ce7ff5ff94058"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-09-03 05:50:02.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cac45646b2c8929b3f5702d9d6269f85c4f0193a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cac45646b2c8929b3f5702d9d6269f85c4f0193a"}]},"branch":"refs/heads/master"},"ae7ea95f5815e2b8c9641e7f0235827c8acba05f":{"kind":"NO_CHANGE","_number":69,"created":"2025-09-03 06:11:02.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/69"}}},"commit":{"parents":[{"commit":"eadfce05d20b72cb547f625cad4fab2031e7e351","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eadfce05d20b72cb547f625cad4fab2031e7e351"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:04:00.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae7ea95f5815e2b8c9641e7f0235827c8acba05f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae7ea95f5815e2b8c9641e7f0235827c8acba05f"}]},"branch":"refs/heads/master"},"575f9fcd55b100d766152187a8eab8134a125d28":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2025-09-04 17:41:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/70"}}},"commit":{"parents":[{"commit":"de99cc14f3d4e297eb85ca98c7107b50ca5deb05","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de99cc14f3d4e297eb85ca98c7107b50ca5deb05"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-04 06:05:21.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/575f9fcd55b100d766152187a8eab8134a125d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/575f9fcd55b100d766152187a8eab8134a125d28"}]},"branch":"refs/heads/master"},"760703c8c7613a4fb31ba9d821929bb61f82f938":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2025-09-08 10:36:04.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/71"}}},"commit":{"parents":[{"commit":"6de44c69fa60e3ead158d77eaf2118bb44435058","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6de44c69fa60e3ead158d77eaf2118bb44435058"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-08 07:56:49.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/760703c8c7613a4fb31ba9d821929bb61f82f938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/760703c8c7613a4fb31ba9d821929bb61f82f938"}]},"branch":"refs/heads/master"},"e664e553fd21c5838f82ac6e9c58d510335ce367":{"kind":"TRIVIAL_REBASE","_number":72,"created":"2025-09-09 10:13:48.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/72"}}},"commit":{"parents":[{"commit":"86a1d9c13c0dc9be5fbf49024a547e47e69f6907","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86a1d9c13c0dc9be5fbf49024a547e47e69f6907"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-09 09:59:38.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e664e553fd21c5838f82ac6e9c58d510335ce367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e664e553fd21c5838f82ac6e9c58d510335ce367"}]},"branch":"refs/heads/master"},"a28260a5e94e8a6ad031b0678f7658479baed758":{"kind":"NO_CHANGE","_number":73,"created":"2025-09-10 08:32:55.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/73"}}},"commit":{"parents":[{"commit":"0c10b3c340c8173a34c8d56e8ea5e0de8e0eb0cd","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c10b3c340c8173a34c8d56e8ea5e0de8e0eb0cd"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-09-09 09:54:41.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a28260a5e94e8a6ad031b0678f7658479baed758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a28260a5e94e8a6ad031b0678f7658479baed758"}]},"branch":"refs/heads/master"},"1f97626d570700e499b6a639779f0f24124486fb":{"kind":"TRIVIAL_REBASE","_number":74,"created":"2025-09-12 06:04:49.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/74"}}},"commit":{"parents":[{"commit":"27b9164eba7d9d164ddf627b6d7c3ea3ce5147d6","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27b9164eba7d9d164ddf627b6d7c3ea3ce5147d6"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-09-12 05:52:51.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f97626d570700e499b6a639779f0f24124486fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f97626d570700e499b6a639779f0f24124486fb"}]},"branch":"refs/heads/master"},"89da923501ecaac95c67fc5b93464f813b069369":{"kind":"TRIVIAL_REBASE","_number":75,"created":"2025-10-09 19:06:48.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/71/938771/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/75"}}},"commit":{"parents":[{"commit":"c0f62a8a42f80a30f3fcfa20b39d6520442e0022","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0f62a8a42f80a30f3fcfa20b39d6520442e0022"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-10-08 07:33:31.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89da923501ecaac95c67fc5b93464f813b069369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89da923501ecaac95c67fc5b93464f813b069369"}]},"branch":"refs/heads/master"},"7dd12c91a618de290ce380d02b458aab79f04e24":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2025-10-16 07:08:57.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/76"}}},"commit":{"parents":[{"commit":"936910938c3798af8a25c7186bc02629c9db5efc","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/936910938c3798af8a25c7186bc02629c9db5efc"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-10-16 06:42:17.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dd12c91a618de290ce380d02b458aab79f04e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7dd12c91a618de290ce380d02b458aab79f04e24"}]},"branch":"refs/heads/master"},"d149436a101683986b1b895bafd4b38c15f431f9":{"kind":"NO_CHANGE","_number":77,"created":"2025-10-17 07:18:43.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/71/938771/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/77"}}},"commit":{"parents":[{"commit":"f240768ff06fd3f1e831918bcb7e734507fe6537","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f240768ff06fd3f1e831918bcb7e734507fe6537"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-10-17 06:22:19.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d149436a101683986b1b895bafd4b38c15f431f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d149436a101683986b1b895bafd4b38c15f431f9"}]},"branch":"refs/heads/master"},"671af175739c651eec0b3394794c1431b9e2ab23":{"kind":"TRIVIAL_REBASE","_number":78,"created":"2025-10-31 06:57:42.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/78"}}},"commit":{"parents":[{"commit":"60ebe49a6f50adc3296dfb247f7ad53e24939172","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60ebe49a6f50adc3296dfb247f7ad53e24939172"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-10-22 12:02:11.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/671af175739c651eec0b3394794c1431b9e2ab23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/671af175739c651eec0b3394794c1431b9e2ab23"}]},"branch":"refs/heads/master"},"3885465dfa35650fbe10220d531b01068010a90c":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2025-10-31 07:07:32.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/79"}}},"commit":{"parents":[{"commit":"dd86fb1156963fb543cc0fa8dab01e2deb1ccbee","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd86fb1156963fb543cc0fa8dab01e2deb1ccbee"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-10-31 07:06:30.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3885465dfa35650fbe10220d531b01068010a90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3885465dfa35650fbe10220d531b01068010a90c"}]},"branch":"refs/heads/master"},"1c1d49419d36667064198e858f401cfbeee640af":{"kind":"TRIVIAL_REBASE","_number":80,"created":"2025-10-31 15:32:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/80"}}},"commit":{"parents":[{"commit":"2e26ba1ad8ae77a01d92f50f6c812d8b275c7982","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e26ba1ad8ae77a01d92f50f6c812d8b275c7982"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-10-31 20:55:40.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c1d49419d36667064198e858f401cfbeee640af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c1d49419d36667064198e858f401cfbeee640af"}]},"branch":"refs/heads/master"},"d3447e4063fbb9c675b42c6ceab853eb33374583":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2025-11-11 02:17:09.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/81"}}},"commit":{"parents":[{"commit":"de06c51b1df8789a53027e7efe3ac1de1e3adeeb","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de06c51b1df8789a53027e7efe3ac1de1e3adeeb"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-11-11 02:17:09.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3447e4063fbb9c675b42c6ceab853eb33374583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3447e4063fbb9c675b42c6ceab853eb33374583"}]},"branch":"refs/heads/master","description":"Rebase"},"ff8497cf2afc0ccb77e0121da5829776bbfd8ae1":{"kind":"NO_CHANGE","_number":82,"created":"2025-11-11 09:10:30.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/82"}}},"commit":{"parents":[{"commit":"c27be81375ddd3ed4dc020cd86a0746aa5d95a46","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27be81375ddd3ed4dc020cd86a0746aa5d95a46"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-11-11 05:55:14.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff8497cf2afc0ccb77e0121da5829776bbfd8ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff8497cf2afc0ccb77e0121da5829776bbfd8ae1"}]},"branch":"refs/heads/master"},"e3b21300e30a72be79b691d970f6511b133361a5":{"kind":"NO_CHANGE","_number":83,"created":"2025-11-11 10:20:43.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/83"}}},"commit":{"parents":[{"commit":"4a3f4ee43a30176fd1a7367ed62e07934cfd2310","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a3f4ee43a30176fd1a7367ed62e07934cfd2310"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-11-11 10:07:41.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3b21300e30a72be79b691d970f6511b133361a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3b21300e30a72be79b691d970f6511b133361a5"}]},"branch":"refs/heads/master"},"08c85e0d85875518941d01060eb9e34e2b6600af":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2025-11-12 09:35:16.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/84"}}},"commit":{"parents":[{"commit":"2aeb2f1239765131a8d3fe74d4b33e67b5d1a57c","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2aeb2f1239765131a8d3fe74d4b33e67b5d1a57c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-12 05:26:07.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08c85e0d85875518941d01060eb9e34e2b6600af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08c85e0d85875518941d01060eb9e34e2b6600af"}]},"branch":"refs/heads/master"},"a079b09f7fa0d2ce312c8471db9f796e7ba6a2ac":{"kind":"TRIVIAL_REBASE","_number":85,"created":"2025-11-13 10:22:02.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/71/938771/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/85"}}},"commit":{"parents":[{"commit":"7838edf6811cf6291c5814d660a00744b5813446","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7838edf6811cf6291c5814d660a00744b5813446"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-11-06 19:13:38.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a079b09f7fa0d2ce312c8471db9f796e7ba6a2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a079b09f7fa0d2ce312c8471db9f796e7ba6a2ac"}]},"branch":"refs/heads/master"},"f5ed2dd67263f555cfb0a721d3e0240f7a1bd1dd":{"kind":"TRIVIAL_REBASE","_number":86,"created":"2025-11-13 11:52:09.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/86"}}},"commit":{"parents":[{"commit":"0d19330ae7c71aa70c6282da92d44c406419554b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d19330ae7c71aa70c6282da92d44c406419554b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-13 05:42:27.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5ed2dd67263f555cfb0a721d3e0240f7a1bd1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5ed2dd67263f555cfb0a721d3e0240f7a1bd1dd"}]},"branch":"refs/heads/master"},"8ab84214404afffb10f2d26cdd55f3e2a0142df1":{"kind":"TRIVIAL_REBASE","_number":87,"created":"2025-11-13 11:57:28.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/87"}}},"commit":{"parents":[{"commit":"112cc7364d635b43afb7bc9c8daf973c4b935dd1","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/112cc7364d635b43afb7bc9c8daf973c4b935dd1"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-11-12 12:19:23.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ab84214404afffb10f2d26cdd55f3e2a0142df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ab84214404afffb10f2d26cdd55f3e2a0142df1"}]},"branch":"refs/heads/master"},"a6584912beb6a09ac1282c4416f7c48434518029":{"kind":"TRIVIAL_REBASE","_number":88,"created":"2025-11-20 05:54:45.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/88"}}},"commit":{"parents":[{"commit":"238689ec9934c7b7cac61cb82dbd4a2e43ccec00","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/238689ec9934c7b7cac61cb82dbd4a2e43ccec00"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-11-20 05:34:45.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6584912beb6a09ac1282c4416f7c48434518029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6584912beb6a09ac1282c4416f7c48434518029"}]},"branch":"refs/heads/master"},"5c7b29a2fb00966aff4277ab1e1b96640771dbdb":{"kind":"TRIVIAL_REBASE","_number":89,"created":"2025-12-04 11:29:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/89"}}},"commit":{"parents":[{"commit":"6466a9a08609271b417d7cf9393c3479415cd301","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6466a9a08609271b417d7cf9393c3479415cd301"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-12-04 11:18:21.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c7b29a2fb00966aff4277ab1e1b96640771dbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c7b29a2fb00966aff4277ab1e1b96640771dbdb"}]},"branch":"refs/heads/master"},"f01815c43a05e97c6f5dc8886d9014a7eba800d1":{"kind":"NO_CHANGE","_number":90,"created":"2025-12-08 09:59:23.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/90"}}},"commit":{"parents":[{"commit":"25702237863867722c646f54dd13caeed8f56f87","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25702237863867722c646f54dd13caeed8f56f87"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-12-08 14:45:43.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f01815c43a05e97c6f5dc8886d9014a7eba800d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f01815c43a05e97c6f5dc8886d9014a7eba800d1"}]},"branch":"refs/heads/master"},"c418eba62de520fca46a591cd195d0eec4d1d6b1":{"kind":"TRIVIAL_REBASE","_number":91,"created":"2026-01-05 08:30:10.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/91"}}},"commit":{"parents":[{"commit":"125c899d073b473de628df403ef3e153a293106f","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/125c899d073b473de628df403ef3e153a293106f"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-20 06:57:41.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c418eba62de520fca46a591cd195d0eec4d1d6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c418eba62de520fca46a591cd195d0eec4d1d6b1"}]},"branch":"refs/heads/master"},"805cab12d6774076265db8e7f0d45f0fb928dff7":{"kind":"TRIVIAL_REBASE","_number":92,"created":"2026-01-06 10:23:31.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/92"}}},"commit":{"parents":[{"commit":"da8a82f85af1a467c0730f87aacfb279887aaa62","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da8a82f85af1a467c0730f87aacfb279887aaa62"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-01-06 08:35:26.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/805cab12d6774076265db8e7f0d45f0fb928dff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/805cab12d6774076265db8e7f0d45f0fb928dff7"}]},"branch":"refs/heads/master"},"f321fd71094f1e56e0277eedc6fc0237ba7d8c1c":{"kind":"TRIVIAL_REBASE","_number":93,"created":"2026-01-14 18:06:26.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/93"}}},"commit":{"parents":[{"commit":"29a9051faa20d52f88bb8533e68ea0cd81187ea3","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29a9051faa20d52f88bb8533e68ea0cd81187ea3"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-14 13:21:22.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f321fd71094f1e56e0277eedc6fc0237ba7d8c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f321fd71094f1e56e0277eedc6fc0237ba7d8c1c"}]},"branch":"refs/heads/master"},"165eba5d70e34d7d2429b29236a8e406a7e4812d":{"kind":"NO_CHANGE","_number":94,"created":"2026-01-20 05:52:50.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/94"}}},"commit":{"parents":[{"commit":"f00a36494e1098031e6c12281662f9a35b47c4d4","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f00a36494e1098031e6c12281662f9a35b47c4d4"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-19 05:52:19.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/165eba5d70e34d7d2429b29236a8e406a7e4812d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/165eba5d70e34d7d2429b29236a8e406a7e4812d"}]},"branch":"refs/heads/master"},"96f5e0fed708fccbecd4d9804984b280610814f0":{"kind":"TRIVIAL_REBASE","_number":95,"created":"2026-01-28 06:59:48.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/71/938771/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/95"}}},"commit":{"parents":[{"commit":"f17ad64cfa5b9d98670a3c014154cfed4c8ba053","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f17ad64cfa5b9d98670a3c014154cfed4c8ba053"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Sandip27","email":"sandip.rajbanshi@ibm.com","date":"2026-01-27 10:57:01.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f5e0fed708fccbecd4d9804984b280610814f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96f5e0fed708fccbecd4d9804984b280610814f0"}]},"branch":"refs/heads/master"},"ad900aab46b7c28873f5862f51066ea2e39a01d1":{"kind":"TRIVIAL_REBASE","_number":96,"created":"2026-01-28 10:38:13.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/96"}}},"commit":{"parents":[{"commit":"1e7b0cd24006385c51d7da9fb335e887443a19f0","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e7b0cd24006385c51d7da9fb335e887443a19f0"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-01-28 10:35:30.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad900aab46b7c28873f5862f51066ea2e39a01d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad900aab46b7c28873f5862f51066ea2e39a01d1"}]},"branch":"refs/heads/master"},"da552ebd3142823777064ef23c8e5c2acd7b1768":{"kind":"NO_CHANGE","_number":97,"created":"2026-01-29 11:51:22.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/97"}}},"commit":{"parents":[{"commit":"5b73b0d942fff652b96c38e20af9201621500c53","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b73b0d942fff652b96c38e20af9201621500c53"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-01-29 11:45:16.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da552ebd3142823777064ef23c8e5c2acd7b1768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da552ebd3142823777064ef23c8e5c2acd7b1768"}]},"branch":"refs/heads/master"},"763e449ebf12ce201a3ab4ca7d1f8a0040b4b661":{"kind":"TRIVIAL_REBASE","_number":98,"created":"2026-02-05 06:38:56.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/98"}}},"commit":{"parents":[{"commit":"1f5a7d3141a202e3463d696b7d6fcbcf00400c67","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f5a7d3141a202e3463d696b7d6fcbcf00400c67"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-05 06:32:55.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/763e449ebf12ce201a3ab4ca7d1f8a0040b4b661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/763e449ebf12ce201a3ab4ca7d1f8a0040b4b661"}]},"branch":"refs/heads/master"},"622d36c3683a1a29466209a836cd2e5a6c760485":{"kind":"TRIVIAL_REBASE","_number":99,"created":"2026-02-05 06:52:25.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/71/938771/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/99"}}},"commit":{"parents":[{"commit":"017dc3f8f9e4ed14ad361f8fbecadba006818f0c","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/017dc3f8f9e4ed14ad361f8fbecadba006818f0c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-05 06:50:40.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/622d36c3683a1a29466209a836cd2e5a6c760485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/622d36c3683a1a29466209a836cd2e5a6c760485"}]},"branch":"refs/heads/master"},"771801fe126dd8e863e2736d7bd0eea4098a787e":{"kind":"TRIVIAL_REBASE","_number":100,"created":"2026-02-05 09:44:17.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/71/938771/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/100"}}},"commit":{"parents":[{"commit":"ebf77dd4cfe3c18c7f3f6974007565783893913b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebf77dd4cfe3c18c7f3f6974007565783893913b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-02-05 09:34:22.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/771801fe126dd8e863e2736d7bd0eea4098a787e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/771801fe126dd8e863e2736d7bd0eea4098a787e"}]},"branch":"refs/heads/master"},"75e756f39570be9a10eb863f562d01e4a48fc9d2":{"kind":"REWORK","_number":101,"created":"2026-02-06 15:01:12.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/101"}}},"commit":{"parents":[{"commit":"ebf77dd4cfe3c18c7f3f6974007565783893913b","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebf77dd4cfe3c18c7f3f6974007565783893913b"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-02-06 15:01:12.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75e756f39570be9a10eb863f562d01e4a48fc9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75e756f39570be9a10eb863f562d01e4a48fc9d2"}]},"branch":"refs/heads/master"},"2ec5a034b38f76b8e31f152313be10dcdb8d6d99":{"kind":"TRIVIAL_REBASE","_number":102,"created":"2026-02-13 09:22:24.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/71/938771/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/102"}}},"commit":{"parents":[{"commit":"4b86c429a3163da02646251246df0ecbbbec8ad8","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b86c429a3163da02646251246df0ecbbbec8ad8"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-13 07:38:38.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ec5a034b38f76b8e31f152313be10dcdb8d6d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ec5a034b38f76b8e31f152313be10dcdb8d6d99"}]},"branch":"refs/heads/master"},"264f490214a792dc185f6224ac6020efd5011628":{"kind":"TRIVIAL_REBASE","_number":103,"created":"2026-02-19 12:24:48.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/71/938771/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/103"}}},"commit":{"parents":[{"commit":"7858383521b33f2ea4202c9f3a607dc24582ef55","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7858383521b33f2ea4202c9f3a607dc24582ef55"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-02-19 10:53:55.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/264f490214a792dc185f6224ac6020efd5011628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/264f490214a792dc185f6224ac6020efd5011628"}]},"branch":"refs/heads/master"},"140164d5c9145d322467955be0f90850870352e0":{"kind":"TRIVIAL_REBASE","_number":104,"created":"2026-02-23 08:23:16.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/104"}}},"commit":{"parents":[{"commit":"75500c24d08d058d8fc8a683f533a3055a4c4be7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75500c24d08d058d8fc8a683f533a3055a4c4be7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-23 06:00:06.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/140164d5c9145d322467955be0f90850870352e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/140164d5c9145d322467955be0f90850870352e0"}]},"branch":"refs/heads/master"},"1bb214071a04111f4acbdf411cb9ecf81a1714e4":{"kind":"REWORK","_number":105,"created":"2026-02-25 15:07:25.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/71/938771/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/105"}}},"commit":{"parents":[{"commit":"75500c24d08d058d8fc8a683f533a3055a4c4be7","subject":"[IBM SVf Driver] Added support for restorefromsnapshot CLI for volumegroups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75500c24d08d058d8fc8a683f533a3055a4c4be7"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-02-25 15:07:24.000000000","tz":0},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bb214071a04111f4acbdf411cb9ecf81a1714e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bb214071a04111f4acbdf411cb9ecf81a1714e4"}]},"branch":"refs/heads/master"},"e6ee5608100230500644e85aea7a813aa69b5ce8":{"kind":"TRIVIAL_REBASE","_number":106,"created":"2026-03-09 09:20:57.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/106"}}},"commit":{"parents":[{"commit":"df837708808e0580039d88e73fdaa3ebcc86ea0c","subject":"Update docs for Pure rebrand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df837708808e0580039d88e73fdaa3ebcc86ea0c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-03-09 07:22:55.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6ee5608100230500644e85aea7a813aa69b5ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6ee5608100230500644e85aea7a813aa69b5ce8"}]},"branch":"refs/heads/master"},"0b619fd008a7b9bc3d127d15e4ed6dbe08d0de31":{"kind":"TRIVIAL_REBASE","_number":107,"created":"2026-03-12 05:41:09.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/107"}}},"commit":{"parents":[{"commit":"c45f193f8eb679e85cd90f9711064af8da95af62","subject":"Merge \"[IBM SVf driver] Doc: Documentation updates for ibm-svf-callhome feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c45f193f8eb679e85cd90f9711064af8da95af62"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-03-12 05:26:59.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b619fd008a7b9bc3d127d15e4ed6dbe08d0de31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b619fd008a7b9bc3d127d15e4ed6dbe08d0de31"}]},"branch":"refs/heads/master"},"3a8dc22373dcaf059a21321b7084cfce762c08ca":{"kind":"TRIVIAL_REBASE","_number":108,"created":"2026-03-17 10:55:03.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/71/938771/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/938771/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/938771/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/938771/108"}}},"commit":{"parents":[{"commit":"efa2e33d53c76497ee02b1d598c005fad893c746","subject":"Merge \"[Pure Storage] Fix performance stats collection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efa2e33d53c76497ee02b1d598c005fad893c746"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-09 14:34:31.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-03-17 10:49:58.000000000","tz":330},"subject":"[IBM SVf Driver] Added support of clone operation for volumegroup","message":"[IBM SVf Driver] Added support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup. Clone operation can be\nperformed using source-volumegroup or snapshot of a source-\nvolumegroup. As the clone volume name is different on the backend\nthan the cinder object, used a decorator to map the name to\nprovider name.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I7ffe7a7f08e7265250d46bd91f654d2a476b8958\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a8dc22373dcaf059a21321b7084cfce762c08ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a8dc22373dcaf059a21321b7084cfce762c08ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
