)]}'
{"id":"openstack%2Fcinder~939154","triplet_id":"openstack%2Fcinder~master~I5caf27edfb690e37a91002285aca88e870b4c77b","project":"openstack/cinder","branch":"master","topic":"bug/2094009","attention_set":{"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2025-12-24 19:20:59.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-01-13 20:03:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5caf27edfb690e37a91002285aca88e870b4c77b","subject":"_process_stack in wsgi disables admin API policy","status":"NEW","created":"2025-01-13 18:54:36.000000000","updated":"2026-01-06 02:52:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e503050c608a15e361a7abc05298106f0648d60","_number":939154,"virtual_id_number":939154,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-06 00:07:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 18:54:43.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-01-13 20:04:07.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-01-13 20:06:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 20:26:17.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-13 20:44:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-01-13 20:45:22.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 20:52:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-13 21:34:05.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 21:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 21:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 22:45:52.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-01-14 02:17:08.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-16 01:46:37.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-01-17 13:15:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-18 21:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2025-03-18 21:28:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-19 18:11:15.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-12 09:58:08.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-12-24 21:31:13.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-05 22:48:22.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"}],"messages":[{"id":"12eeeb9211b4ecdca5dc5eb39fff76928c625333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-01-13 18:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9ac2950fe2a73bbaa97fbb008262a66b5524b49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 18:54:43.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1195/","accounts_in_message":[],"_revision_number":1},{"id":"421591148a68e11ed53517a0a5baa3ca143fc07b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-01-13 18:56:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"63408a567545508b34e809078fae196c8b907fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-01-13 19:02:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b84d849af1b31ad824bb7e0f3ae001bcefad46a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-01-13 19:21:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dbf51f9cdd158b8024ae9ec0f15e96604e1ad380","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 19:22:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"4cb99e98945bd1f680dae7a365c6017c9ff0eba4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 19:22:13.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1196/","accounts_in_message":[],"_revision_number":2},{"id":"b4d474fd50d4255b4479f17d96208f9fcc3a1eee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 19:50:21.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"6c8bab3596653746fcf0c3e92aa3ca20924776fc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 19:50:22.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1197/","accounts_in_message":[],"_revision_number":3},{"id":"5011a6793498a21020e2008ee2a511c981d5222c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 20:03:38.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"e3a88f04abd6c5d054b1b4d1f9bb5eaaafb4eb34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 20:03:40.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1198/","accounts_in_message":[],"_revision_number":4},{"id":"59239e679b593386479bc767c77b09666e50922e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-13 20:04:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":4},{"id":"a3f6376ed643dc07ba8c78e81911e503437e1730","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-13 20:06:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":4},{"id":"38b375249287f0126007a535fb025be64615acdb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-13 20:17:03.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"1207e59a8ca5c02a6f685c6a38df5237f3cb6e76","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-13 20:26:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":4},{"id":"c3ec1031ba03842d8c35ca105f0fead2553358a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-13 20:44:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/222/.","accounts_in_message":[],"_revision_number":4},{"id":"c6d8499e85585b4adf73f6c5b36f14b3412274e1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-13 20:45:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/939154/4/ : FAILURE in 1h 16m 01s","accounts_in_message":[],"_revision_number":4},{"id":"0dfa4c2adb538c5b1e6f998d68b8b3f080b83edd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-13 20:52:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2a245f8dc36040cfac34a3e188c7666a\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42e0cd16f71541e38abd33053f22ca16 : SUCCESS in 39m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4a72399098a64db9b365a8d108a03f0b : SUCCESS in 44m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e14b7d6e8a384cebaa63c0c2d157b0bc : FAILURE in 1h 30m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9bfdd2b78924e8aa458f85d04664afb : SUCCESS in 39m 14s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd3958aa75ac4fc19f3cfd355bf90e51 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2a2e99a035f4968b369779095ddc883 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34b8dd03888a492aaf5ef23687dcbf40 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2150581465f24572ab58277e4a1ebdb9 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"38e1588e23e253ba754a2679f21b414311dfbf9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-13 21:20:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/211/.","accounts_in_message":[],"_revision_number":4},{"id":"6dd9f88ee4568b806a6dcf414f2a5eb93cae4bdc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-13 21:34:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/af34afa16e27405ca828dbccfaf1102a : SUCCESS in 1h 11m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2dc65dc31f9542768906dba2fc1d3374 : FAILURE in 1h 20m 44s","accounts_in_message":[],"_revision_number":4},{"id":"e5475453e42bf6264bfd20a668bb07a1edc962f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 21:40:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84890750013947d19628e61d2ca58125\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2dfaffc22fde463491eac3d8b609034a : FAILURE in 24m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/826273e489ef4f2ea3232c720fb6dfb2 : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c50ec13f1ca2ad76e1f212643c2ce6cea19378b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 21:47:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/674464cd69a14554ae45bd084bde40db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c102be51e1444ad78f7229850126f7c4 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab46470a04ff4874945629b8de66fc5f : FAILURE in 13m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6040198148c14601870e7a40f98dc4d6 : FAILURE in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca008a74c754f1b867711bc10135dbc : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/33175e9568b742a497e9ec0c6d97c8dd : SUCCESS in 47m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/258809230a344beeacd520be4992e3be : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c46749aef7d24623b0e70515d2bd5100 : SUCCESS in 1h 16m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba13d2c447194e609f1fc0fc97f4a0f8 : SUCCESS in 50m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/09855344442248b1ac55057fdc59ecfb : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a57fcbffae824f19983c96f7f21215c8 : SUCCESS in 7m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e74dd24e1caa4e4686b429a97a4d7f99 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bdf4d02a81844a22a8b5c976af00dde0 : FAILURE in 5m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/07febb321d19487a9b87bf766d30ebb4 : FAILURE in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4769498828b4e199be3aa316ab384c1 : FAILURE in 35m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b93d3c82b8514cdb9eb450a897d0a934 : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e1363677a0c4e3692ba2945431eaa42 : SUCCESS in 1h 47m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d68e773356cf4bc182b1d63d981fb7bf : SUCCESS in 2h 23m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/977c6831644445c0af42a82e1c03e4fc : SUCCESS in 2h 02m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/90ced6bf1a5c4bc4b1657895d251c687 : SUCCESS in 2h 20m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c208c8732978455a98febcd6c0e22afe : SUCCESS in 37m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/05ba129fb96749489b3c1a99e94cb775 : SUCCESS in 56m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c53da6261970464e96a8867438abdd78 : SUCCESS in 1h 37m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4a82baafa57b4c53a9e41e2be3421da8 : SUCCESS in 55m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f1026a376d2847978c268e27b01ee608 : SUCCESS in 1h 03m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0beb457895241c7a7380407ffcdd632 : SUCCESS in 1h 13m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a17d49797924b6e8248a64b94e59219 : SUCCESS in 1h 47m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab61ffc743b94cbb929a51ec139184eb : SUCCESS in 1h 15m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eae2b0e429fe40908e8929aac1eebe26 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":4},{"id":"492795b74956098e3edc34bf6f30099fade1b2d3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-13 22:45:52.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-3/FC/ : UNSTABLE in 1m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-3/iSCSI/ : FAILURE in 3h 43m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"68c0f84bec0b5703e4dcc363202ed3741c5232b2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-14 02:17:08.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/939154/4/check/fujitsu-eternus-dx-fc/88d0276 : SUCCESS in 2h 38m 14s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/939154/4/check/fujitsu-eternus-dx-iscsi/ebab9f4 : SUCCESS in 2h 57m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fb8b03dff8f109f0887f87d93ad31239748ce3c0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-14 03:41:30.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-1/FC/ : FAILURE in 8h 46m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-1/iSCSI/ : UNSTABLE in 2h 45m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a3f57116b894fc33448b39cbc75c490b62b0368","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-14 03:55:00.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-4/FC/ : FAILURE in 8h 33m 21s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-4/iSCSI/ : FAILURE in 3h 10m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"64b2333ff6a9585fecbf2df38e0d037c2d426b22","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-16 01:46:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/4/20250116064447/NEC-V-ISCSI : SUCCESS in 3h 59m 01s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/4/20250116064447/NEC-V-FC : SUCCESS in 4h 01m 47s","accounts_in_message":[],"_revision_number":4},{"id":"6cd49d8923d2c171ee41b9dfba813a098940dbd6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-17 13:15:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/4/20250117140919/NEC-FC : SUCCESS in 5h 57m 38s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/4/20250117140919/NEC-ISCSI : SUCCESS in 8h 06m 04s","accounts_in_message":[],"_revision_number":4},{"id":"9bdfe863f7198f358293ae5bfdab78ead288f788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-01-23 20:02:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"de25f6b8992c48889e148f0e2913bc16843767b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-23 21:19:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":5},{"id":"b026dcce94d46d99569a75e213852b2ac8d1dfe5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-23 21:40:16.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/24003f0f3cfa43c480bc4f939e10b060\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d33519146ba84872b1eb44a1e03e2130 : FAILURE in 38m 49s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aade8ff93f6f4108a011f0035a64d966 : SUCCESS in 52m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9a487a1c18a49548dd272b45b4f7d20 : FAILURE in 1h 37m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85c920ac251941acae71d26fa3c25c86 : FAILURE in 46m 07s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b6f6b53caf94a7fa16570ec10d997c9 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fe9c96ee3cb413a80b94b9f9c51eaba : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4365f1ae15de4da38d13a689f9446b1f : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/677cc2b482674ba688d365d9f851c4ac : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"787e3e3495efc58206479e19e4fd2ec9d76c412d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 21:41:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6f5d10946743448fe6007d224a5150\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99632b492dc145a3ae3a69fef931f84c : FAILURE in 31m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/866aaed11ac04a25b191bb728f16db46 : FAILURE in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9fb3c498917f1ac58af110177818a77ee55b782c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-23 21:56:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":5},{"id":"085c0bb6219cab6aa7065ac021d2c3364bcbf9a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-23 22:33:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e4b6b3793d084211a5f3864be69963cd : SUCCESS in 1h 16m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/41bd4add2f3a4b73b8c266f9b2e5fba8 : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":5},{"id":"20b7f8792be69102ce614c5740b4e4bf60133fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 22:52:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c243847d035140ccba8cddeac3fede6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/184b85b618cb4e54a8dc6e5321f5c36a : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be554500d8c6403a9551801002dcbfe6 : FAILURE in 12m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/266b1ff584ce4fddbac3f0b23abe58a0 : FAILURE in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79cc56741c84f22b7fd2bab53ef009b : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/2767fe2df704438bbf1ee043558c8170 : SUCCESS in 1h 05m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e25063d669404e2fa9160415f7d1a191 : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1eeba03396624bb28d0334674f7b2cbe : SUCCESS in 1h 23m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5552ca479a054e188a85d87de8ac2dc6 : SUCCESS in 51m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1678aa21f0e479080df45bee0a3ce7d : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f245e41a38594608b0a3de15baa75fb3 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e17ab5d1bd54ac087c953313571cc6a : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67ccd3d5d7254d11a832311a3e827853 : SUCCESS in 5m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9b05fbe142154f03be6d21d3008ab66a : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d5cf19167294cb5ba57cef06aec533b : FAILURE in 34m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ca2d81a297d4693b53e57000c193519 : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e6c879c5c144b34a9a2af980d382bcb : SUCCESS in 1h 44m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/39610bfe194a47eb8abfaeedda371f9d : SUCCESS in 2h 41m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a881a1f93b27440294dbb89aadfd43a1 : SUCCESS in 2h 02m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d32cb5726868404f9b263b3660fd2cbd : SUCCESS in 2h 05m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dfac14ed0d22424fb1fbb3aa96807480 : SUCCESS in 27m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1acf7b5e236b48d4b877c496eff2351d : SUCCESS in 1h 08m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df5a8e3a8d6b40919a84543f2f591aee : SUCCESS in 1h 23m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86af02d4148341dd80543d2f934d9a70 : SUCCESS in 31m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8103abc82f64d2abe23cdb43e3f7c20 : SUCCESS in 1h 16m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/868dfe3370454527b4f83d1a30df69e2 : SUCCESS in 1h 03m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce21a016a1094976a7ca2344667d87bd : SUCCESS in 58m 37s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/555b068c716c4fbfada9fbbd8523171d : SUCCESS in 1h 32m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c2216e9f75c4bfdae04d333c623ebbd : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":5},{"id":"7e2b95825bdba6c9599f95795175f81f6edc00a1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-23 22:57:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/939154/5/ : FAILURE in 1h 27m 41s","accounts_in_message":[],"_revision_number":5},{"id":"023721474b725a541f461862caa801a968e50684","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-24 10:59:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/5/20250124152535/NEC-V-ISCSI : SUCCESS in 4h 00m 09s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/5/20250124152535/NEC-V-FC : SUCCESS in 4h 33m 54s","accounts_in_message":[],"_revision_number":5},{"id":"15b145cf41d01568bc22de0710cdc7cc26755552","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-25 00:44:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/291/.","accounts_in_message":[],"_revision_number":5},{"id":"be856a62035000c6675a96da514692f676d07db7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-25 04:12:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/273/.","accounts_in_message":[],"_revision_number":5},{"id":"dbee9c96faa7667842380c982af325fc293fa754","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-25 04:34:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/5/20250125051632/NEC-FC : SUCCESS in 5h 54m 53s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/5/20250125051632/NEC-ISCSI : SUCCESS in 8h 18m 16s","accounts_in_message":[],"_revision_number":5},{"id":"60461a3195df21686518e1170e288c62bded3567","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:51:32.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1259/","accounts_in_message":[],"_revision_number":5},{"id":"b627498db26c95a04b94d19e2aeadc50abb6235c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:51:34.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"1f5048a68668931f559e5748f85c55b68559157c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 16:22:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":5},{"id":"a0efedcdeec4525e26674e928e2f80c4f9fbae04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-02-10 22:14:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"16cf4c7acbbb49b068718996643686a4269fbe77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-10 22:14:58.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1317/","accounts_in_message":[],"_revision_number":6},{"id":"3b349b9b3370ce54a31cd751b2cad3b9a6d100e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-10 22:15:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"cfb4307855b6e4ec40fd76a406e122347f63cc38","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-10 22:28:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":6},{"id":"cc05e90023bebd64820112f985ecd428002fc1a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:40:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679cc5a258f7425a852ced375e4b2214\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da7f7c2e859842cda0112609b8fd8512 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e77513ec26c4f43aca23a8b5ef9f7d8 : FAILURE in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b46f8188c7262ba7fa1dd37cc3ac03a17af9bd0b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-10 22:55:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : FAILURE in 39m 10s","accounts_in_message":[],"_revision_number":6},{"id":"ab3576f55f92bb25987af37e4597997e45cadfa0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-10 23:15:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/357/.","accounts_in_message":[],"_revision_number":6},{"id":"5e5e25276bd02238335133751adc152654df2928","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-10 23:35:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 19m 14s","accounts_in_message":[],"_revision_number":6},{"id":"17e8aeae6acb6ab1bd6c0b4141e5e290dc1b80e9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-10 23:45:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c5f12f3d43054a0f9f2d797530998d9f : SUCCESS in 1h 19m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/28bf375f851f4a6fb515fc6fd834dc35 : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":6},{"id":"8a1b1b2ca05bc8fbce72a34e24ab4ee9c5fcb7fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-10 23:55:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0302b3787a984ad0a62f2b3ba530d6a5\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f8f0ebf08bc84297bd42d523f08d32be : SUCCESS in 43m 33s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37618de527bd437fbade9b6e65618400 : SUCCESS in 38m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5315cb1a684467489ed3026113c15ac : FAILURE in 1h 40m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/538103ae30f84f01b3da6f44a0d0096b : SUCCESS in 46m 07s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2df0e52b568f419883fd7cc8b2bda769 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d0606a23830454194bd68039c9f532c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b201a128078f4c568f04801c705a1dd5 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36054b39510344559d9d9f384f18b292 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"1a96211b4a1dd267ff3591bd29e88609adbc752e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-02-10 23:57:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/939154/6/ : FAILURE in 1h 33m 59s","accounts_in_message":[],"_revision_number":6},{"id":"d5c2037e82ad771d09ba3dd89d6a8809fdab77ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-10 23:59:26.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/339/.","accounts_in_message":[],"_revision_number":6},{"id":"58a8e88f8d1d8620def0802259c47fe7890982f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 00:25:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/364c0874b9474b2c8ce7f0ab74167617\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69645721a374751a8c034412321ea5b : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/178903a87b48442d99754a10afd54c14 : FAILURE in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b340a7bd21544ebca0371e666cac1ce8 : FAILURE in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e77cc50522c4fe78728095acd9a7177 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/0d00de1253464959a6ec3b7ae21c7157 : SUCCESS in 58m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3d7331acb264790bf94a29f0352372a : SUCCESS in 44m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a1cd3c5b4f140588289bd4cc2e70aa3 : SUCCESS in 1h 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2a8bacf62e404b83853cb736a655c4 : SUCCESS in 1h 04m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/38ddd3a030934aa5bc57ddf40fb03099 : FAILURE in 18m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99e77d2492bd47b9a82cc3273e75d59c : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d73118e174d24675884151da8ca3b92b : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a42178f8a6c48658dd7f25578a902d8 : SUCCESS in 5m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/af9ed7d013174d9692186eac328ecf9d : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/df6a214ef7de4f679ba68b246fac46b4 : FAILURE in 29m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/632432ad93fc4c6890e7b7cf477d8631 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e9610670d224187bd7133cb77e8d0ed : SUCCESS in 2h 04m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1059a4157948455a8f350618f30fd52c : FAILURE in 1h 57m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c07c31bde34e4191a0380401921a9a76 : SUCCESS in 1h 43m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5bdf1b2e516f43dfb7b9eafe2bb369ab : SUCCESS in 1h 55m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/41a1600979ec45b8aaa87ddf4119f8aa : SUCCESS in 33m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c38300aa3dac46e5ab88261fd8cf9470 : SUCCESS in 1h 11m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f112979358d746fab3cc28ccc7e57fb2 : SUCCESS in 1h 38m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54f46f018622474ab58b4bdabb7db7e8 : SUCCESS in 50m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/659f0537af8f4c779e9e4b8854b1371b : SUCCESS in 1h 48m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5e473d721c1f412da01794bc9bccf20d : SUCCESS in 1h 47m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/594bf3b2d90f41cd961383f21e1f0baf : SUCCESS in 1h 30m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6282dd5278c4656bd7325b66e4f0087 : SUCCESS in 1h 49m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c8287b8c4d34a769e19548955483da3 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":6},{"id":"691ef83bcc6d76eb0a003d5cbda35d20193c7445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-02-11 20:06:37.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f0b227304a86e56cfa1baf1e95e5def5e1e1d407","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-11 20:06:46.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1320/","accounts_in_message":[],"_revision_number":7},{"id":"2bd58285d136cda24e8000c14d21bb61778949e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-11 20:06:48.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"f2c0e0d84431766fca6ed6f937bb75f1ae24069c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-11 20:20:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":7},{"id":"6927de4a9802b2a35e11369de973eda4a4c3ec1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 20:36:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cabda4489f846e4b7386c59f1820f9e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d127b2e5b34c425ca38caf0ff5916b1d : FAILURE in 23m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f8dfa168b0148ae99514037fc911d1b : FAILURE in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"80b9e2d124ed21bb2d33213da30fcfde19a74453","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-11 20:50:09.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":7},{"id":"a58d86e2c2d6165aab8f0f1f868c0bd02126d7e5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-11 21:27:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":7},{"id":"e1ba0eb443c08c9b9cb1434010162cfd8dc796ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-11 21:28:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/342/.","accounts_in_message":[],"_revision_number":7},{"id":"c5c58538bc7c9e215d53e8eb2850da5129e855eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-11 21:29:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dcd9ba62c0434529942923871e55e523 : SUCCESS in 1h 12m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/733f9d4af0e2411bb6424bc513c071ba : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":7},{"id":"748da051fd24e3ba04da6267571a55f507e4837e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-11 21:43:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0b3d24fcd4144481999a8a65d74f54ab\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0238510b63c24a82aa5d8de9793576d8 : SUCCESS in 39m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30f4f152ad434a3b9e86a0cae1278fa3 : SUCCESS in 43m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20e5c3b3aceb44ab8c6d66ba34733317 : FAILURE in 1h 36m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5e2c8c1d91aa4d27b44b206214a0d63e : SUCCESS in 37m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f4b7ffaa4f9f4ed1840704675758d618 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/abb7dc01cf1448199f91bf66272afb57 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/753f34a5bb27429aab6eac9be812c8e5 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8244ee68e8e4b459a10385560154f73 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"e6d296231a6072b71bcae9aa706c58bccd8008e9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-11 22:13:39.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-7/FC/ : SUCCESS in 2h 6m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-7/iSCSI/ : SUCCESS in 1h 27m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"a83b2e18e80560975b7a711fd3c898444df93bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:55:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d1b0cd87aa34ba68d2d7452deab325c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b822b63e7c9541aca55ab6587af3edb6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbddcf7394be4ab7ab759f0354568ebb : FAILURE in 8m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bafe97e393cb4c4bb3d6ff41c3e43a51 : FAILURE in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9763d6ed5940493d9f901b37e6c0f116 : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e41d7d543bff4d6bb3abad9a7199b00d : SUCCESS in 1h 12m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cbcb358d23f4f05a594b9c591335c3e : SUCCESS in 52m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/edda0d5dbfe8461c91412a9b16509e57 : SUCCESS in 1h 49m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbabbc7d8f5c4917b3c1b4a790138e6c : SUCCESS in 51m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad02319749ab44778b236f902d73d04e : TIMED_OUT in 40m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2280ce3065ce4e8a90f9b6e73e7049f1 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00d19d9945ef48d7b58a857cd7221062 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4793dee355504e9f984c422a87d7ab3c : SUCCESS in 5m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/804324fec7c84b36bc3e81293bb93cec : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3822af1650984ce08cb6d85364fdaba5 : FAILURE in 34m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90226e2961e449489baab34b1bef77d6 : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3dda2c3ac55e4771b17eb96a70e20372 : SUCCESS in 1h 27m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/47be979b6c454343b219db6e8197f9ba : FAILURE in 2h 46m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7de4b8821fe74a74ba8558791c27902d : SUCCESS in 1h 21m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e822ab3ddaa6419b8fc99233d782c16a : SUCCESS in 2h 16m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b38f6acbac64fb09f33cfc80de2307b : SUCCESS in 21m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7962fa0b524d4e6fa89528bd57dbc30c : SUCCESS in 1h 07m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91200597ca4e48b6afff6d35dfbea7bb : SUCCESS in 1h 20m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a91f43119e2b4375aa179b99daf786e3 : SUCCESS in 29m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c30694b94a644e2bd58f9e2e16b442f : POST_FAILURE in 2h 15m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2b347be4845746ed968d0589051b2d48 : SUCCESS in 1h 53m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ab02c6bced445b5b446022d1fea57ff : SUCCESS in 1h 36m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c633a2f6fa504a3abdaec5277d7fe51d : SUCCESS in 1h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e6f633160ea44708768c8276dfdf320 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":7},{"id":"b5a717fa11e25b8d09789e55e3f862fdaed2ee8a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-12 17:55:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/7/20250212183842/NEC-V-FC : SUCCESS in 4h 24m 57s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/7/20250212183842/NEC-V-ISCSI : SUCCESS in 8h 17m 12s","accounts_in_message":[],"_revision_number":7},{"id":"9175061d85f911f834922ede4b28a0777cb1fa80","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-12 21:20:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/7/20250213002721/NEC-ISCSI : SUCCESS in 5h 25m 22s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/7/20250213002721/NEC-FC : SUCCESS in 5h 53m 15s","accounts_in_message":[],"_revision_number":7},{"id":"a30dd50f4cf2c7af31cb0e86b22c5bd8cd0f1334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-03-12 21:51:56.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6156311f7ac55a3928aaf5643b7a61efb6e4a8d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-12 21:52:08.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1466/","accounts_in_message":[],"_revision_number":8},{"id":"cf7c92c0401545a25ea8e13b13bbe4cb7f615990","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-12 21:52:10.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"bcaf853dc5e89796017586681982ab7c1e836900","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 22:10:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4664c49b58564009b8b385d647fc24b2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ca2f9e26524435192ef7f91e1f7829f : FAILURE in 16m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a310da5ef19945e3b2478c196c0f4553 : FAILURE in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1db327c722d2b5f6147bee3341a27f7203313a48","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-12 22:11:48.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/396/.","accounts_in_message":[],"_revision_number":8},{"id":"ac74140fd8763cfe5ad058ee04e7bfd123521ee2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-12 22:27:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/506e62a7f716405a86f137ef9a8eb46f : RETRY_LIMIT in 5m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/83988f4211c24f6191b9e5cfe67d5bdc : RETRY_LIMIT in 5m 31s","accounts_in_message":[],"_revision_number":8},{"id":"fb9b36987487a197474fe0b271078c6011c40f37","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-12 23:07:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":8},{"id":"257c748333016bfc1b33f869df8310b79f50cfa0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-12 23:34:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 04m 32s","accounts_in_message":[],"_revision_number":8},{"id":"1b741931cf833502796d9ff8fa5978888819252b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-12 23:49:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f19fd12a7c3845349aca619c759e8416\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7dcef1a993b4804bab5592d515ad0f1 : SUCCESS in 42m 41s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d804bb9013114d29a2aaef2a3714376e : NODE_FAILURE Node request 200-0000029136 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fed35e81d0d04060adee0a0e625550a2 : FAILURE in 1h 57m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1fd570e1e82432eb3e9b85e7722e8f5 : SUCCESS in 40m 38s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/63f05f315fa1418ab7a8da7b28322c88 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/89c8f031c00546f3883b33ebba03903e : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/300e120a54ae46f99277dc85b1dcae9d : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3364a3f17b646aca216437d74158686 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"cb2a9b53f1b1883439911e1fd157fa0ddeb1314c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 00:36:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab2a22e9b6a64aedb5a1b9ae201848a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a536988a38f04d80a51ede61636e9096 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a69d444916433eac24370a7bbdf0a4 : FAILURE in 11m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b6f7538287a46a3914e4d1322c7f03e : FAILURE in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84cc697c98394fe78b5214eafe058118 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/217c45e4f7f44a769cd5d2c94f1f57e6 : SUCCESS in 1h 21m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8197fe72cd634e0cb1aa52b94b2d3cc6 : SUCCESS in 57m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a92711fefe8241b69ada7fcdf84cad5d : SUCCESS in 1h 24m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/447c72e1823f4e8e92c36d97d360a622 : SUCCESS in 1h 07m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84cf851cc36a424e811c545f31db98e1 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/26390b8532b845debbaff77fdb665876 : SUCCESS in 2m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c53bee425ec44bba966167205698121 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f798b4728f304548aa684a87747a6362 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9135bafc985942e0856727f787b7039a : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c4cad67824ff486a8b972c9194ea22d4 : FAILURE in 35m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30aef0feb3344c8a87b5e67c75d53520 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3aa6fbba9c1c41bba945cb30d811ab45 : SUCCESS in 1h 52m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5051b78c715c4073a7aaa317a856ab99 : FAILURE in 2h 38m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bfae4c3efff64dde9b792c2b857870b3 : SUCCESS in 1h 53m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1d210b0114684f148d993eda411c3f02 : SUCCESS in 1h 54m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/77b1fd98feba4a7ca4bc2fec6be0af98 : SUCCESS in 23m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdc6b0ba2758403491c0230761290a8a : SUCCESS in 1h 06m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ed87fc726f34245b572c7acb8737b04 : SUCCESS in 1h 26m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b9e1bf36a264802a57b69258e30e1ea : SUCCESS in 1h 03m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a42768e3fa294689abe055994c3bf717 : SUCCESS in 1h 31m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/11bc0af4215f49719ece021ebc262b47 : SUCCESS in 1h 38m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ceb1b53e9484e2aa2ff5055d36e2cb9 : SUCCESS in 1h 31m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/133f29017095413db33332b6675238c6 : SUCCESS in 1h 23m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/976eac461da84220937bef011a5fee84 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":8},{"id":"2833124b8d63efa94924fdccf9d61b5ba709653e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-13 05:20:45.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"c62bf745e5e76be2c5e7a1ec97b618c499e71f2e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-15 14:23:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/8/20250315191622/NEC-V-ISCSI : SUCCESS in 3h 53m 21s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/8/20250315191622/NEC-V-FC : SUCCESS in 4h 07m 26s","accounts_in_message":[],"_revision_number":8},{"id":"79180fea804346b59333d457f8e599c355464881","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-15 19:45:45.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/8/20250315225113/NEC-ISCSI : SUCCESS in 5h 28m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/8/20250315225113/NEC-FC : SUCCESS in 5h 54m 29s","accounts_in_message":[],"_revision_number":8},{"id":"a24d5db655528005132e4cec9ed794ac9b01d4f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-03-18 17:59:53.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"41f022433cae107b7c6491f2866a568ce96e94e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-18 18:00:04.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1517/","accounts_in_message":[],"_revision_number":9},{"id":"402d500bd37712b9052aacb682463dbab07e4b4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-18 18:00:06.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"1cda59dc352d12f502f8c658065de649ede5c289","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-18 18:07:01.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/523/.","accounts_in_message":[],"_revision_number":9},{"id":"f1091d501cb1a369f7ba0f899fa47fc983fbba25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 18:17:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32db00dd44c4bcba32507fc4301b674\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e622fe3a1b4a41d1a60aa05729b3502d : FAILURE in 15m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71309faebe3e41e69e9f3e6bf2ad5b75 : FAILURE in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"22ec8c6f4a9c0224d6b92e57fda4164d30c29426","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-18 18:20:27.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3e6243a45b6d4c65bc52a58a200a7934 : RETRY_LIMIT in 5m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ef91b618d2354df790099c0a3535584d : RETRY_LIMIT in 6m 08s","accounts_in_message":[],"_revision_number":9},{"id":"08e544e52d92530e378cb9c23b3c7fc8da41d8c9","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-18 18:39:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : FAILURE in 38m 21s","accounts_in_message":[],"_revision_number":9},{"id":"80344bbf43c9fe052d60c15bfb0bbc8fada6f75c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-18 18:43:44.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":9},{"id":"b8bcb6e908abb98810c3f47bc3b4d5a359d2c8a9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-18 19:03:17.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":9},{"id":"b45eb239a8a540b251f8edba594d35b94c36582a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-18 19:03:49.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/454/.","accounts_in_message":[],"_revision_number":9},{"id":"7976d166e912ecd4974c14d95cf56c3ab9feb467","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-18 19:56:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/da7cdc25c28f438d9f4aca59d73b05b6\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39b363106690487faa3edff3c378a8ea : SUCCESS in 40m 37s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4450cd69130941368b18dad7034bfadc : FAILURE in 52m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a142c51b165d4565aa6b65df8a6f5709 : FAILURE in 1h 55m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9c5090db22d4648839bd19b33454ba8 : SUCCESS in 39m 37s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80e7927940c149d19491e9bbe00e0691 : RETRY_LIMIT in 31s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f6898e1accc462284cebfa413024b3e : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90a3d07353254d8695336ce3fac56e7b : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fab15188914f4083952a051f53d3f301 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"0c0a8e5a64bb3ee284637699ab7f3467f3a289f3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-18 20:13:51.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-9/FC/ : SUCCESS in 1h 57m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-9/iSCSI/ : SUCCESS in 2h 13m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"81e6e482b3fee4af0c8a48680f7e2f85726dfc22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 20:51:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f48cba49ea94b3a8e702f460fd7dd55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3add185fac74541a10424681840e999 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb82f0c7c0cd4f8f8a19528358d3d9be : FAILURE in 13m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27a3f925ba45428ca71cdc687afa8dac : FAILURE in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab969210f6d4312aaa7373d01b0aee6 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/08f14f872ddf48f2b9de905587551c8d : SUCCESS in 31m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73cdd8ca8ccc4f9696b0f6c2b1075a6a : SUCCESS in 51m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7ce24a3f3e04a45877770b738f42916 : SUCCESS in 1h 29m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ee3832aee064969afb1594a54fdc700 : SUCCESS in 37m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86e66c5823314115ad6eb4f482adcd68 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/01bc568846984b01a2b9aec835a6a7e8 : SUCCESS in 2m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73de209b67c748d3a03b48101caeb65f : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c56398413cc4a8ebf548a01954a6550 : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ffab9f164d5b42548929c6a28717b305 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/141e817a8da544a0a16d05577ed54274 : FAILURE in 35m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b19d89c42874a6c88f02038198a13e5 : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/de8cde69022f4ddba7b9c69bc8fe74bb : FAILURE in 1h 54m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2739f13397e14e33a5c136f58c9ba561 : FAILURE in 2h 51m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/53d168f234d647418d632f54b4815092 : SUCCESS in 1h 41m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/73909991e6a4454db09ff03c0fd7bb29 : SUCCESS in 1h 56m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6dd62b38bce04f98ab478476839a477e : SUCCESS in 25m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/67f05f7f11e7425b88e70d5892a57c86 : SUCCESS in 1h 05m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29a7236435fc4742bdc8d56555a20513 : SUCCESS in 1h 21m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7485d4d7dd64699b685dd358ff5495c : SUCCESS in 29m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3f0415035614defa439c3efbbcef26f : SUCCESS in 1h 16m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/33250f3c836447ae86f9930dfd7b8959 : SUCCESS in 1h 18m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8211e61bfa9d4fd4ac550bd8cbb37107 : SUCCESS in 1h 30m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07686a82dcb1472596384ed3421c8624 : SUCCESS in 1h 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca912cc9631b46bdb81abe9144021b51 : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":9},{"id":"16aaf35dcee83bb0141a9ef51e59eb010b58ca7b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-03-18 21:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/939154/9/fc : SUCCESS in 0h 49m 03s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/939154/9/iscsi : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":9},{"id":"203636d000fe861810ce3277d306eb5a13e8adf1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-18 21:28:47.000000000","message":"Patch Set 9:\n\nBuild Successful","accounts_in_message":[],"_revision_number":9},{"id":"077cb4ef205c9d84dbc429a26e53f8ecea74d5b1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-19 06:07:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/9/20250319103346/NEC-V-FC : SUCCESS in 4h 07m 33s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/9/20250319103346/NEC-V-ISCSI : SUCCESS in 4h 33m 36s","accounts_in_message":[],"_revision_number":9},{"id":"ff66739ef13d3d4bfa56a2bb3b904277125b43d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-03-20 17:09:45.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"780ad7c1de0211ed97f495392afcb05b062ef857","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-20 17:09:54.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1528/","accounts_in_message":[],"_revision_number":10},{"id":"075202aa31b7f37eb3dabf32a0f620ba35ecaf56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-20 17:09:57.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"6d350f3823656878d0e75012df635b64789f551c","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-20 17:10:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"df4324d47f1060af23428b34341791bb06b498fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-20 17:17:20.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/465/.","accounts_in_message":[],"_revision_number":10},{"id":"ff57dfa52fed0d612b2a33577e111030596cf3c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:31:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4909ca2b39d48b78be9c88bdb825fd9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64902ef2e9594564bf3ec922e7f23014 : FAILURE in 11m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/684969b2092c4841b43cbb9766b3bed6 : FAILURE in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7be59fa7dd58e9602dbf61609915a6699a437a56","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-20 17:53:34.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":10},{"id":"265c8ea4fff2735df652a87f8358ab5dc97f925d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-20 18:11:16.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":10},{"id":"65abbe8424d028330670e033f3b636a9bdb39885","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-20 18:21:51.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/534/.","accounts_in_message":[],"_revision_number":10},{"id":"538ed8501e2f3149d9ca2fe5bb866473703af7e8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-20 18:57:09.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cefda4d3913940a99917355ea27f0ed2\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f72d5cb240f94f0b8c94825503400528 : SUCCESS in 40m 20s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4647a89386a24d2e965451dbec0fd8f0 : FAILURE in 1h 24m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c5e40ca1a9b47cea2bb49ecb5f75fd5 : FAILURE in 1h 46m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b4a7e9100344be3b355f963fc35c1ea : SUCCESS in 40m 12s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e42aa5595f8d4156ad804bca136b187c : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7fc62466d1b474595966e7c563a5374 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8f0cd767d9143ec9c769d6f11e38e80 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4523ed4b076417888d412ff42345ec1 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devs... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"1bba1619a588391fc0c7bbcbb135d367a1fac7dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-20 19:40:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/69e43e985a9e4d3f9f764e07dffee1d9 : SUCCESS in 1h 14m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/60e0310cd2d84717a13c8d148512106d : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":10},{"id":"318332045df66f21b4be4ef74aba261a8bae0bf9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-20 19:46:32.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-10/FC/ : SUCCESS in 1h 56m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-10/iSCSI/ : SUCCESS in 2h 36m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"f60e38c8a753216b624b9b39f7f86947bf82abcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:55:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f7fe5f5ab7540f18d9ea5b0778b0c5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984517874fad430080a1635fbac2de2f : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c0797852e9e4cbda116112044c7017e : FAILURE in 13m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4a787e1ef2e4651bd8e466e93d0a69c : FAILURE in 12m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d4fa09584f4a3cacfec2002de6bf1a : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/e134448f1984455eabc3491e9bb1db38 : SUCCESS in 29m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c916c6263e8d4f34a2d71a4912af5683 : SUCCESS in 50m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95e0f847be8e452a918f0bb2771015eb : SUCCESS in 1h 28m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/923ac718469c4e95a216d363c97d6602 : SUCCESS in 58m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c9d232e0d184577bfab84db34aebb1c : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/922a4f097d2b44b38d442468b97d4518 : SUCCESS in 7m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0a7564dfcd1466e9b41e7aa3e80378e : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9d3bbede6f0f45f99375b11215055b28 : SUCCESS in 9m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9eb8a914cde74fbeadcaa5587cde77aa : SUCCESS in 28m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e7019811983047279359bc5f4d21aa40 : FAILURE in 38m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afaead94f0504ce4a4a48dd42c6be88c : SUCCESS in 7m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d391993d09544a98a8d19641a06d6391 : SUCCESS in 1h 10m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b19e6fef41b416f804a603ca39839d7 : FAILURE in 1h 33m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fa126cab8de4c69a5e2fc165c59fa7e : SUCCESS in 1h 46m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/05009235d9984d589aa188486b4e9f1d : SUCCESS in 1h 49m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/add08cb0af8949d99de068aa3b4b9de8 : SUCCESS in 32m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee67e81369454487a4e00f3346cc2281 : SUCCESS in 1h 07m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa3e7563b4494206ad8ae054cb35e2a4 : SUCCESS in 1h 24m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2e41135cd19249a8a16bc011c88c7d16 : SUCCESS in 38m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87bfa521973c4934af0bc7475e98e6ad : SUCCESS in 1h 27m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0449ee7ff5c497d8eb654dca1a0e2ce : SUCCESS in 1h 42m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16122c2730c24130910057e214f9c99a : SUCCESS in 1h 28m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cdb2097f87d4b6d936f37b100ee8ff2 : SUCCESS in 1h 21m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ea164890bf48adbb6c6bdfbc337b1a : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":10},{"id":"9fe59517803c2daee03e80aa07160f2fdf67a3b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-20 20:52:40.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"22cc7e289865e7f46a153b830322c384d5e8511b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-03-21 02:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/939154/10/fc : SUCCESS in 0h 51m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/939154/10/iscsi : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":10},{"id":"48a457eb79cb096d9affd106a5855765e40e2e45","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-21 10:37:34.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/10/20250321151452/NEC-V-ISCSI : SUCCESS in 4h 07m 00s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/10/20250321151452/NEC-V-FC : SUCCESS in 4h 22m 39s","accounts_in_message":[],"_revision_number":10},{"id":"e242b28988b90a95cd85e6bc29eba81ee330abe0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-23 08:37:23.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/10/20250323084821/NEC-ISCSI : SUCCESS in 5h 28m 23s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/10/20250323084821/NEC-FC : SUCCESS in 8h 48m 58s","accounts_in_message":[],"_revision_number":10},{"id":"b79e98a9593c774744859f705cff7e8a13bb4c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-03-27 18:34:07.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6d5c8e9780c1983e5bad27b40ebb672b44e6531f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 18:34:14.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1608/","accounts_in_message":[],"_revision_number":11},{"id":"6309fc863206c435df47c84887fbe969074f11db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 18:34:16.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"395cfdd8ee58cc74ee1a02e8b8e224b54b07d6ef","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-27 18:35:08.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"0866b793ddd9ecbfbd0c25d7439c040b5107b423","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 18:51:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415331d1176d487c9e05dfca2846dc0d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1629b15068604b999f958825b359445d : FAILURE in 12m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b440e1654bc42b48764af2a366ecbde : FAILURE in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3d47320b1e72bfca413eb8fdbd0f23cd1796b63f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-27 19:15:06.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 39m 28s","accounts_in_message":[],"_revision_number":11},{"id":"62c7c1af66add16d8ddcfeaefc7983f3135eb722","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-27 19:20:43.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":11},{"id":"df195a3645af1e93b6b054b48c0aa27ff5e572d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 19:25:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/511/.","accounts_in_message":[],"_revision_number":11},{"id":"9c14393bd104d8ceb9c8c0dbfe3f88a6c717b3ab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 19:53:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/596/.","accounts_in_message":[],"_revision_number":11},{"id":"5ee0bc742de45364c7edc5bde5de91281482ccbe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-27 20:04:57.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-11/FC/ : FAILURE in 1h 30m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-11/iSCSI/ : UNSTABLE in 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"4d1b4be692150c91aa6945ddb65dcefd5882df98","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-27 20:25:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c798cbf26edf476f969f5696c67f1564\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/382e65058e664df48d013b8ea52c52ac : SUCCESS in 41m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d47ff649b234851bd0a3a5142dbe91d : SUCCESS in 42m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c967eab8d6b34518b9464954e754a60c : FAILURE in 1h 50m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/885ef774d2364ba5a954e927c155c7c9 : SUCCESS in 46m 11s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d2502ce8cd24994a2bf7b3fcb0a0595 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/68d264d024a147a9a10f50cae8808635 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7dac74d2080b4729b8a907302aea4d4a : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57a04b88e5e24c55915ff3a855a85257 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"dec62c650abf33803a9783138211c02020c01148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 20:53:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02383247bbe046ddb2b6ad4b0a3e532f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8394c7ea7e4199ae6e8ea1ec9ba46f : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0c39498e814a759dc9682cfe43f4d2 : FAILURE in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b1c4caf45046f4838851a0feb9ed2d : FAILURE in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409185e5a6f54f299efe70cb7884ec08 : SUCCESS in 16m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/45732b06a3374662a2cd6abf43c3665b : SUCCESS in 49m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cab60ff539f345e8ab0ec27d12482bdd : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8004d988ceff49ffa082adb73195eb9d : SUCCESS in 1h 38m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc0e00e76b2441bcb05be88ce1a77c1e : SUCCESS in 40m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da00dba0a57a478f8c834f9a651e4a4e : FAILURE in 23m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2338fd0666c449d184a22d1811f9a10d : SUCCESS in 2m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5047651a69c741fabf4ea903ab7ebdef : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44cbe1049b6f4d70a385bcbc341a9023 : SUCCESS in 3m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93ae45ad78ad4d8a98fbbb664d92ed03 : SUCCESS in 4m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d23d883ff86b40b099cbac147835009b : FAILURE in 17m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bffe7185aeb44ccb9d969bf6e09f0803 : SUCCESS in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/455b1b6d062d4d218a6a0ac4b9cdf9c8 : SUCCESS in 2h 05m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e72c64c4e275415e95dff779edebd9c5 : SUCCESS in 2h 13m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7126effca9cb4d96ba7651b9f14120d5 : SUCCESS in 2h 02m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eec8f202afb0434c9fa51eb346aa6c08 : FAILURE in 2h 05m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8841ba9782c8488e85a5e58f11b3d0f8 : SUCCESS in 33m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/754dd026c0ff4e5b8566ccfefa117117 : SUCCESS in 1h 00m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fcd52fea93454884a0288278a4280c10 : SUCCESS in 1h 17m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2396cee565444a5ebbd1174f16e8493f : SUCCESS in 45m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/943702258025400d93e2572fc7e51f96 : SUCCESS in 1h 15m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/000f1ee54d204379a220628801c3da45 : SUCCESS in 1h 40m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f52e341215744e2da5b336f79cf44ec5 : SUCCESS in 1h 42m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f248973f8eb842a0906f5798cd96db5a : SUCCESS in 1h 28m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d59111946754e29ab5f4a91023a3915 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":11},{"id":"8dd0c56bf856ddd26ecb8fbde7bfa06886eb224c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-28 02:32:04.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/70b0fc8370074943b2ed99a7a657cd1c : SUCCESS in 1h 19m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/33ac90e43df3480d9a74137b040b9d5f : SUCCESS in 1h 23m 34s","accounts_in_message":[],"_revision_number":11},{"id":"53b9db9828592e83cac9ba7df7687d64ccc0cdb3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-28 03:37:04.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"49b711f08d6cd2d3f4d833872ce7e708dc5b80d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-04-01 17:11:48.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"165c217d32c159e83f71f6a235efe7192179b35c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-01 17:11:58.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1633/","accounts_in_message":[],"_revision_number":12},{"id":"1f1c297f925c37c702914ed63f4e8971444f7d37","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-01 17:12:00.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"c035866e751197d04fb9b3495e700b3688555855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-04-01 17:12:34.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0fcad3ad8ad85dc9a3c063c608eaa031a67e5857","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-01 17:12:43.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1634/","accounts_in_message":[],"_revision_number":13},{"id":"298654af1bf9a73c40602ff9d26a5cc8dd7a1e0e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-01 17:12:43.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"0e33c3ed578941d5a7acbd02ba6537043e0d6c60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-01 17:12:44.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939154/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"c9d299de8be3e8f16a03317627e4960531ae852d","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-01 17:13:23.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"60aca29d2e41bbd14b0d97903727793d2f0f1de9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 17:33:32.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff3036308ab47489cf3198354629131\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f6d0ab135634286a3fa865ed45831ca : FAILURE in 17m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d724e6f2a034eeca1c132f33f3327f4 : FAILURE in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fbf080893ff8fecf14c0ec6e6e8349dd2eee6f38","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-01 18:06:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":13},{"id":"bd635fa56aa00f6d1e955daa05832e31f176d1fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-01 18:16:58.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":13},{"id":"9bc60660fd6502c7e5a17f5bbbaa15e34f9dc14b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-01 18:32:15.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/624344622cea40dca11684a0e32f4eae : SUCCESS in 1h 14m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a7bb0c6fa84e4a82b15ba8653d960ef3 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":13},{"id":"f4b9db0df90bab54f00d48ffd23d079945ddaa49","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-01 19:04:43.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/531/.","accounts_in_message":[],"_revision_number":13},{"id":"95562e030ae27a891b05e625a0b4d751209c8ee1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-01 19:08:49.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/96598095dfd74541abb44b2b897be8a1\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19ca7dffdad140529f24f1e515e64752 : FAILURE in 48m 51s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b679924e1624374826c96e300663ea8 : SUCCESS in 41m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9937791097ba4617885a83682cd54712 : FAILURE in 1h 51m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/977e60d9cc94467aad4e7cd815d89806 : SUCCESS in 46m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33810e8cf350498ea84fdb2b22caf0c9 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed2d9d1aef694c8fb822a920d86b5dcb : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5cc292f4dfe42fab0c9d36f714d7c90 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa8b03d2b3544fce90662260d0d98d4c : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-d... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"b69851b875bcfb4f592b5e368505215eea105ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 19:43:51.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf8a2d5b478245f999f7f90d820499c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daf90265048461197dfb72b53063fba : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcb092699f5f4f60a80993b9dd7c93f2 : FAILURE in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364a1db39a9e4d3bb5acd7fff6039a42 : FAILURE in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55562e77ccb4ccc83640577df578e9f : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/e489a9ff7050492daf54cac24a9fb726 : SUCCESS in 27m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/651186574f38493d8793fc9ce3861399 : SUCCESS in 54m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55c4d590fb96474fa4f023f3e623fab7 : SUCCESS in 1h 22m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/952ab7a6a9584ed3b764b9a9ce3621c0 : SUCCESS in 1h 04m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1557fec08274dc2bb11cf45d012d012 : FAILURE in 20m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6179264384744d50a60cbf6d832d25bc : SUCCESS in 7m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c20a11c2202448038ae91abc361c8508 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3897717118324a73829d64bc400ed050 : SUCCESS in 8m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16e786528b334b0991e8c7cd35977cd4 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dbfdd17e55c44588865dd07454bd8d23 : FAILURE in 29m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb7dcf6e282e4aa388b4b6e432e78a7d : SUCCESS in 7m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4742810fbcbc4df69ddc6fb49bcfb32e : SUCCESS in 1h 50m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/906c7288d4fe4039b3a0c075f45865e1 : FAILURE in 2h 28m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6ed8e4e2b7c84badbc1e445316fd9bc4 : SUCCESS in 1h 43m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e2d4bba29dc14c24a3b747501aebfcf1 : SUCCESS in 1h 30m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8a454b77598e4974942d41d08875192c : SUCCESS in 22m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3de0b0a89fed4b298ff615d80afd5866 : SUCCESS in 50m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dc36d123b60a4c11bb4f1f24731c419f : SUCCESS in 1h 17m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef44abcdda614c30ad08668a0aec9c71 : SUCCESS in 28m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7f601253a9c4689ba7b3282f6b6597e : POST_FAILURE in 2h 08m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4ef11d32a923474297644d89baea4721 : SUCCESS in 1h 24m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3c17a3c9bd94a4b97237eb63d66433e : SUCCESS in 1h 44m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f40131055b346bc9ef5504f4de15bd0 : SUCCESS in 1h 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d979368d51174663b069e27ec3199e93 : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":13},{"id":"5011185ecbdd5fa091fddb6d6a352c0617a5ee9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-01 19:55:49.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/615/.","accounts_in_message":[],"_revision_number":13},{"id":"3cfff7b3703fc428b6bda332330245aa9bb17542","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-04 18:37:28.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/13/20250404232048/NEC-V-ISCSI : SUCCESS in 4h 09m 18s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/13/20250404232048/NEC-V-FC : SUCCESS in 4h 16m 36s","accounts_in_message":[],"_revision_number":13},{"id":"1ae4f4cf815418d8ace3bd978728ae730ac84da1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-06 11:20:21.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/13/20250406112505/NEC-ISCSI : SUCCESS in 5h 27m 18s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/13/20250406112505/NEC-FC : SUCCESS in 8h 55m 13s","accounts_in_message":[],"_revision_number":13},{"id":"1d05aeb0e9b28c099a11064404988c93560b9d7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-04-22 18:41:07.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"5958627a8b451b78b1619c2677f1fb5d05c358e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:11:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d15100ba416a4113ac75323a7adc67f5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/765d616e7b9547b59a352314b25a438e : FAILURE in 10m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e96355d5d9b43e09119b6d13bdc7e27 : FAILURE in 22m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a13fc75dcd2a4cc799eeff52157bf9c7 : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"55852a120909875c41c33504023ea53a678108b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-22 19:47:28.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/11a82d71dcbe48c5aa2e9adf371fd026 : RETRY_LIMIT in 14m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5d71cd2daa2b43e994c4f81f98dfa2a6 : RETRY_LIMIT in 16m 23s","accounts_in_message":[],"_revision_number":14},{"id":"2f26eb9412f3f1fa2ad9baaa2770fe696083bb14","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-22 20:09:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939154 : FAILURE in 41m 58s","accounts_in_message":[],"_revision_number":14},{"id":"3ffb7b95753990ba94ea4333a4a01df7bbb21cab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-22 20:22:51.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/94329e86e8524e11a1e15d85545c3ed5\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef6ea1b8399a4281963099df8fdb196b : SUCCESS in 50m 28s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b92e58de1c84a1c9a6d13be15672bf4 : NODE_FAILURE Node request 200-0000039191 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/49942877e3b0447fb4c167013803b601 : FAILURE in 1h 38m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce42525fc38742e487008b765673a4bc : SUCCESS in 44m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f184d150a924725afdf5382f0c5fb0e : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ce58d2eada84d69bd52170382b79911 : NODE_FAILURE Node request 200-0000039195 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7332351fd97746ed9cde2f6b8cdfa866 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34f7c0f87e4f40768c9f2685be87ec8f : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":14},{"id":"7ff9f08ca092cd0a9096bbed2401ee7f1816989c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-22 20:25:23.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":14},{"id":"c122c13926fa849cd2932e7c88b73451ec59e284","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 20:27:48.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/750/.","accounts_in_message":[],"_revision_number":14},{"id":"42662346a9971d9e77bb14f7feb408ea4530a86a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 20:28:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/631/.","accounts_in_message":[],"_revision_number":14},{"id":"a60755b9e9a232ddde3e25af7150684424fb1f33","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-22 20:50:20.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":14},{"id":"c62082269624a4d6f1b482aba4828015a80eacc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 21:14:32.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8f9540fe0924cff8a10e75503eb4117\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938007a713b443ef860d0118aaf87c70 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b5b1ac23b524500866fe560af0547cb : FAILURE in 8m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/227b6cc062564c449704e23cb65822a5 : FAILURE in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e56b47aa33f44e5b913c71eeddf21d32 : FAILURE in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d695c9ed23554f74895a6a20e9381976 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/cea39f2973ad40a69ed63178daf96adc : SUCCESS in 52m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e81a43b875a4eecb2bdf2325cda6431 : SUCCESS in 32m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4716ce16dc9b401f8609dfbf8ee60399 : SUCCESS in 1h 22m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78f3846386f345718c0108ac5c657af7 : SUCCESS in 47m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a2a864555f274046a2b48670738dc6a7 : FAILURE in 15m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15208ff9a98440298ee07c4af28d2652 : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/613b3bd9b0ee483d82393f623e82081e : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9619e568e4204efa93651ba85c69eba4 : SUCCESS in 4m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a8488e6ccaf349ba958b9043bbadafb8 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8c6519abe4e46fa880ffcd7df505ff1 : FAILURE in 24m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd0dcdea53fe460998e16f21ffcb45ff : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/808f6165b5dc4bd197cfd12594643ac5 : SUCCESS in 1h 31m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d0559db718b34b25a3591ab6d0391e18 : SUCCESS in 2h 32m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65b87408e6474d39b3b624dc4f13bbd0 : SUCCESS in 1h 15m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/932e0690f8bd4af1953c02f3009018e4 : SUCCESS in 2h 02m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/563710d40a124ef0a6e028ca4844ff12 : SUCCESS in 33m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/211330b2b707480db467ab5fca85b4ad : SUCCESS in 58m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82294642957147a6ac53ac46c07d2ed8 : SUCCESS in 1h 08m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4fe1c5c2f63c4e00b9c8376ce65bd796 : SUCCESS in 40m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d980b306bff240108ccc2ee3d96673cf : SUCCESS in 1h 39m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a3c27ba05ebf4e34bbd521bfceea871c : POST_FAILURE in 2h 17m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/162539b693d84d80b1d36dbe9a553cf3 : SUCCESS in 1h 36m 27s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2567ac7842294dbb8220738e9c8cd592 : SUCCESS in 1h 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/183cc44034d54fc68dc1fa5ad3f8d06f : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":14},{"id":"c08e707b9409f85a9698e050462ff0cfaceb2c36","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-23 09:54:52.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/14/20250423143056/NEC-V-ISCSI : SUCCESS in 4h 13m 09s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/14/20250423143056/NEC-V-FC : SUCCESS in 4h 23m 54s","accounts_in_message":[],"_revision_number":14},{"id":"b2cc613647a39362ae4b131ea5d37ace97d91292","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-24 00:11:50.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/14/20250424004509/NEC-FC : SUCCESS in 5h 57m 21s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/14/20250424004509/NEC-ISCSI : SUCCESS in 8h 26m 38s","accounts_in_message":[],"_revision_number":14},{"id":"f64d1fdc0a2a7215a162dd850642d2e9f78e5360","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 17:57:16.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"928250516cb86ce0c310690d97ffab23c584a339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-06-02 21:06:10.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"f0764d05982f41333cafdabce7f55055809ef3fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:31:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e541c472beba46b0a16805e71fb7d851\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/762d2d849ff4413e987e4d80ac6ff05b : FAILURE in 13m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0573016cf3a54be4b2778d215c151d72 : FAILURE in 14m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0789cb84d82b433689e56c76a68f2be6 : FAILURE in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d9d6a7eed454749226a0c17599efd4a0dc5aa873","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-02 22:19:13.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"dadbce63cb7a508d9862b07a153c672fb3b3823f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-02 22:40:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":15},{"id":"a05d21f10fa3dcc9d37280886cb803bed28c3b46","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-02 22:42:52.000000000","message":"Patch Set 15:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-15/FC/ : UNSTABLE in 2m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-15/iSCSI/ : FAILURE in 1h 36m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"7a6cf4cb661e27d7d20f315d178fd500f348cca3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-02 22:43:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/812/.","accounts_in_message":[],"_revision_number":15},{"id":"cedf1018344062b2ca0402f7c17250bcb22ee27e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-02 22:47:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/936/.","accounts_in_message":[],"_revision_number":15},{"id":"d06a862560ad428b4632b7248778100a030ae918","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-02 23:42:02.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 20m 04s","accounts_in_message":[],"_revision_number":15},{"id":"274b4efcc81b19d2216dd4c489039860935c9999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 23:47:57.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5aaae202cb844da59193cf60a56e5b07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a22696594943d8b6ac30ccf1d63da1 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc501de611314ef288eee74426e5e09a : FAILURE in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c756b74671f460da1f4e451db16c4c0 : FAILURE in 9m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc50ecc269204f689edc50a866fe657e : FAILURE in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666659fa7190480d86ea5220ffd8537e : SUCCESS in 15m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f46da54cf3374115bf092778619f2dd9 : SUCCESS in 57m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b23bc27511a42218bd06b804b34ef2a : SUCCESS in 58m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a37e7158f63841e79c652ebef050b7f4 : SUCCESS in 1h 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72e38291198e43229eb44f53b6d1a536 : SUCCESS in 55m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc91334a39c541adb5429e0ee05f8686 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ff04a4c8761d4a5db0e89776ff62919a : SUCCESS in 7m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c6e962b47f0f4d028caad777b4c19295 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66f3acf3e82944909468807c9ec19d5c : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/79a62eb518fd47b092f84f2aaea17d16 : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b6d32f865d8425c866f6974ea7f2c8b : FAILURE in 40m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ddce3774cc844e5a3ae20324747c971 : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f462ea039c4c4cc1938a4fdda41dbfc8 : SUCCESS in 2h 00m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e69ae0c6707441ef833c51b6e1e0d6dc : SUCCESS in 2h 01m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9221bef5d0c4e1d96df3523278d82ee : SUCCESS in 1h 55m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7b71e72935d4277a12b5eaa68fbec40 : FAILURE in 25m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d07d906db3434329a5745a6ff510c82f : SUCCESS in 26m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a5e27651919446bb21e4bdcf77e9e53 : SUCCESS in 1h 06m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b0af17647964304829101258e111a25 : SUCCESS in 1h 12m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b9aac9da28534d66a3f71767bb253ebf : SUCCESS in 47m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b09649437a944dca9e2449205fee264b : POST_FAILURE in 2h 29m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9c6522f5044b4a318d6c867e4dfb56ef : FAILURE in 18m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f30aa361e5f4ef799f17c63f8e2eafe : SUCCESS in 1h 23m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e043d6d905b4f19953fb54c01cc088f : SUCCESS in 1h 47m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ed2b9cb2bc549e2943853b88b9d9d62 : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":15},{"id":"4081ab5da90ef40460bd01237fa9b04575815397","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-03 00:04:55.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/31921e20acf64ab2874792248fc08549 : SUCCESS in 1h 14m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2e7e8ab27c8a4e7ba6664c5d49ec4450 : FAILURE in 1h 37m 28s","accounts_in_message":[],"_revision_number":15},{"id":"0eace7532df3481dd87225ef4b8cd116e5d04304","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-03 21:57:16.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/15/20250604023945/NEC-V-FC : SUCCESS in 4h 11m 05s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/15/20250604023945/NEC-V-ISCSI : SUCCESS in 4h 17m 28s","accounts_in_message":[],"_revision_number":15},{"id":"d0bd55b94c0370310f7d48aeed34766b2cac23d2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-04 03:08:27.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/15/20250604055927/NEC-ISCSI : SUCCESS in 5h 45m 26s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/15/20250604055927/NEC-FC : SUCCESS in 6h 08m 57s","accounts_in_message":[],"_revision_number":15},{"id":"eb9751f880165fc0b98a1b5e08db838f749c98ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-07-08 20:54:02.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"84507c01616dfbbb7f18628d2b94938416062f16","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 20:54:49.000000000","message":"Patch Set 16:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":16},{"id":"6be34293254fd6331a4f397ca91d953022f0d5db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 21:13:56.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe029a03905c4413971b4d2abb0eea22\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bb9cfa19c0248b8bf506e0ac003f7dd : FAILURE in 15m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b1150d08b374423a4bc3fb09ad012e5 : FAILURE in 18m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21b9a307bd004be192fedb3b5b1a2fa4 : FAILURE in 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b4373cab1d58380156dcdac111fb316a252a41ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-08 21:57:43.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939154 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":16},{"id":"d517f0f22078274ec9c3e3d82f246b58666dd025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 22:47:34.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96dd91bc19d44e7886bc5d9ebdfa82d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d47d3900b7643318403dd9a3d37c0b5 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/110adf9cda3f4415ab5212ee14c2cb61 : FAILURE in 9m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c3e43b0716b434fac6f3e08fa5e0eac : FAILURE in 10m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/633d9142755c4e4cb4277c1d46570961 : FAILURE in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69d16d897784bbba67f157edadc03f2 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/857ee922a42f4e83b1298753466801e1 : SUCCESS in 46m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8239283dc3814553a62511493614f1ce : SUCCESS in 49m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/710dc98b33a04c5dac9d196a4914728d : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a496ad82b2c49e0bceafd432ba3c0d4 : SUCCESS in 49m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/44f4bf6700ce4b408a963631898958d2 : FAILURE in 22m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98c4db7a846842a1aaff222bf224cc1e : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5091e77b27e64b7e8010c7dff7684a58 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30099b6007ad419585e4c9eebc85cbc9 : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ed1b43d7bf534ff2a742c6fd9f6f7e7b : SUCCESS in 8m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4b7bd9c3fcb34fee96f904dd3a0109c6 : FAILURE in 28m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e7a3387fb8c44d3801a7849f323f5ca : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/290a769333564d9498015c768b3c263b : SUCCESS in 1h 42m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ecf632d113b94b7ab93b2cc752d837a5 : FAILURE in 1h 22m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fdcd23edf92444e68eee410645689a64 : SUCCESS in 1h 48m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/54eeb5d9119d4ab7b8bcd99d4c731e5f : FAILURE in 21m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9b07a5d570f74188b75f49095153924b : SUCCESS in 17m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15759403cd4047d0ad614997be8855f8 : SUCCESS in 1h 06m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/618bb03d093f43fab484eddc1d30eb9c : SUCCESS in 1h 32m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb8671e0ec644c34823c54fb15fa5d1f : SUCCESS in 30m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0fefb9bfb65144de98d3ae06a9074d46 : SUCCESS in 56m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ba73d66359134293934456b7e19a6743 : FAILURE in 10m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34649f6dbc0f43569a3880ef8cf3d4da : SUCCESS in 50m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9630bc6e10624491832f05e49cb0b1db : SUCCESS in 1h 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e54b88e150ef4bee9f930a9a3993724b : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":16},{"id":"e2d57f5bcd37fa7aae61e7a08bc758f21980c811","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 22:58:29.000000000","message":"Patch Set 16:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1197/.","accounts_in_message":[],"_revision_number":16},{"id":"206636db2f69223ad415a2fa665f4e77e924301b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-08 23:14:35.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":16},{"id":"9dca1e0617ef0191df9490c816a95030ac41fe1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 23:39:29.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/edcf4db1728f43dbb2ce0f89d997d794\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/733886646c784710abdae539ebe721d6 : NODE_FAILURE Node request 200-0000072342 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41259a9ebf704e1a97950f4acbf417b2 : NODE_FAILURE Node request 200-0000072343 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3857f85b9fc2493ba104409d13e58a28 : RETRY_LIMIT Host unreachable in 53m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a007c66a97c24a2db4b44a404527353c : RETRY_LIMIT Host unreachable in 53m 05s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e50696d52d73449ebe3747717acfce87 : RETRY_LIMIT Host unreachable in 56m 51s","accounts_in_message":[],"_revision_number":16},{"id":"a0d7c4ce2638ee58fb57f445ad3cf8b943750c66","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-09 00:03:01.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1327/.","accounts_in_message":[],"_revision_number":16},{"id":"d7c28b6beebd3673e3b2a3f39937b6b0ae7630de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-09 08:24:28.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"e9a975c06511372beb8f0d83c347dda1902d6b07","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-16 10:01:29.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/16/20250716133332/NEC-V-ISCSI : SUCCESS in 5h 05m 43s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/16/20250716133332/NEC-V-FC : SUCCESS in 5h 27m 55s","accounts_in_message":[],"_revision_number":16},{"id":"ce50ee367173712ae61ca32ad02ea585a22bf5eb","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-11-13 18:10:55.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"977f91ac8524a7652a1cd81f37e8079fa5ab4c01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 19:03:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b051d4143274f1d94f74e0c52f17a95\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/935df4a413fb4eaf9083a1df29837376 : FAILURE in 21m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd5bcc0b6a1049a7943824567f67f275 : FAILURE in 22m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b6cc597f5ab43e69026b2ad79854ab5 : FAILURE in 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8f567732492b38d9aefb79be3f5841ac117bf4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 20:47:52.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba2c7ba728f47f09bc2bf577dec3ed7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57e39ebd72746658af96107610f3d3c : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f20764e6cbb40639ae6eeb572e14b4b : FAILURE in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a62aec2e5e54ca297d5c788b54d3704 : FAILURE in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776a49cc53ad4c1abff724a4b081c400 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/bd841caeb231431caa370e9ce26c5513 : SUCCESS in 35m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f445676a43484ea0af3b45b6f0d9c36d : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab535fedce934b5fbf18b6d6d4dba627 : SUCCESS in 1h 30m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d9a10437b3b45d49e07679d8617a9fa : SUCCESS in 34m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f4b621e287745f88619d2efd1304199 : FAILURE in 24m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c431f83a780c4f889e979910e7db5b1d : SUCCESS in 3m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76eb18e0dbf24301ae6f5f22d1218753 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4bc153be841d461898b8246142679d4c : SUCCESS in 5m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7a06df0f832548a79c2ffc0d0624684f : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b72481d825d3445caed852f27e87c598 : FAILURE in 23m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43a56c0ed133434e9a249b1b6f9321e9 : SUCCESS in 3m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3c59cc67dd24c96ae269953558e93de : SUCCESS in 1h 47m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/47af5cdc52704815a208ccd9cbd68a43 : FAILURE in 1h 45m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1fcb705ebcd443f3b2e0a60e51e43cd9 : SUCCESS in 2h 15m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/884f832e75564328adac9979ae738d3f : SUCCESS in 1h 55m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f6f248c0925049fabf27e903e5ef2775 : SUCCESS in 35m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5def25608f56433f91a4a3bdcd8baf20 : SUCCESS in 1h 10m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b957f9eff1f49a1b2de46ce38724339 : SUCCESS in 1h 33m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aeea93f4db784a1fbb7c8938eb38fb7e : SUCCESS in 37m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a76ba7bd4e649cfa37521a481e90105 : SUCCESS in 1h 58m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/de9d311a6c7b46a68c92f1d25ddb73f6 : FAILURE in 2h 30m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/561dadaeb4ea4d81a6b1a1c0dd973a18 : SUCCESS in 1h 15m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/375713f95cd34f3bb836e48763060f78 : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":16},{"id":"b3b9509f5f0155d0bb5cb1adb9d3ab7e921231b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-11-13 21:22:00.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"157bfcb8280902d5754e44ada1f2a7ef800479a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 21:22:50.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":17},{"id":"afb198ee5de2fea68a4b4cfaae034ddf6805cd0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 21:47:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/809fe994fd7943aeadb8d2fd84c9f205\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b86f55a620b6456f83c3c640aaed4cf6 : FAILURE in 16m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5de235288f0e4dc38e7293054577c84a : FAILURE in 18m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fdb9a6e3bf114d2db67dd8f71471dd80 : FAILURE in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d161ef38078dc347912c8b131d32212572d04a51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 22:20:27.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2479/.","accounts_in_message":[],"_revision_number":17},{"id":"d301d2a55ae49ba1294a1d96b7ae1cfe6599726f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 22:21:46.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2593/.","accounts_in_message":[],"_revision_number":17},{"id":"ebf3ed30e8f3b795e2d1bd593b48b80e4b3af125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 23:40:46.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bbf62d6c4ff4997a6e7bf46fae7dac1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fdf8977d634f7397227f1b31a9f218 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d793d341b3f642049ec2c32f5a4f18f1 : FAILURE in 12m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73b3373028344f17ad8aa1f79c6be509 : FAILURE in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04ee4fe9a254a608b79245727cdfc18 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/3c743cd6ead44eae9d74d5c1047b50fa : SUCCESS in 41m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9616bc26b622426fb2d6e7f4d01c2046 : SUCCESS in 57m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/629b461f228b4a0bbe1f312d07a2dd32 : SUCCESS in 1h 23m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2c295b1df7446a09c5b936bfcf38ff1 : SUCCESS in 44m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/db364144fbc741c68512ab60dc11b1b1 : TIMED_OUT in 40m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/37a1d646a4bb4fea86fbe69b80d89308 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a20e7ec8bbaf4806a9beb9726c6b30f4 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/297d7b22d1ee4ece9cf9bd1bd97da68b : SUCCESS in 3m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/490b1adf81aa48f080dcbce7d65687a4 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2326d9fdfff2459ca764ed60ac818476 : FAILURE in 40m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/705fbb5d4baf4ab08611fa74130ca825 : SUCCESS in 3m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e8f03340d16436695a5ab8ed702ed10 : SUCCESS in 1h 50m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f6be4ce1993040e7b5c1f390039952bf : FAILURE in 1h 47m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0adb4e0cb1b348d5867b9487818acae8 : SUCCESS in 2h 05m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d333e25d29234195ab520f433bb91737 : SUCCESS in 1h 46m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7159de22aa7441ba9e954a134bba66e8 : SUCCESS in 19m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d6fee253745a481c99cd3d06976398b6 : SUCCESS in 48m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fdc36a6a7c384f139996d59efaf27ad1 : SUCCESS in 1h 36m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/94dbd01b2056400486b2283e62a73df9 : SUCCESS in 25m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/61257a278bab42018b46fd5cb169bddb : SUCCESS in 2h 12m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3920667b99714e96909d38ff9a139268 : SUCCESS in 1h 07m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43ea2c411d614f98a71d4962570baee9 : SUCCESS in 2h 05m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecb3e7ef227c4f7a853f2b6b82d43bcb : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":17},{"id":"06b791c7f4a291fc9ce8a836654630c7f965a606","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-14 05:39:06.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 13m 42s","accounts_in_message":[],"_revision_number":17},{"id":"c23e01185b4cfff1fbf0198f74b3947d3ec59966","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-14 08:28:26.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/366cdeb2e8804dec82f55050509ae9c7 : SUCCESS in 52m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5b063a16c60341b5a22e4fa2140676c6 : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":17},{"id":"926a7ab4d09a2c996b9fde039df628bf7a6865c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-11-17 18:57:55.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"397ffce37e3fc041dda75a367d785a1fa6a4c1ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-17 18:58:46.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":18},{"id":"e53a45ccd801266b69c7c580c1a0877db40af2e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:16:39.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9a6768b2b9b44b3b58712ad70d10bed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a82823153104f69951872cccf6ede76 : FAILURE in 11m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e665674b5fa4a3e9751cfa34dc39c5c : FAILURE in 16m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c38ad90cbdd49329aff4d72361694fa : FAILURE in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8f4360ab9bf73a82fcc37fb752a056ebc10e1268","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-17 20:10:29.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":18},{"id":"8c7b568ea8b2e930742310e9903a09ff09b6c065","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-17 20:28:13.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 10m 12s","accounts_in_message":[],"_revision_number":18},{"id":"0b70e8c8fee4e864d4dcb9da2b34b933478a63d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-17 20:39:38.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a3f338ef48724cf384d2990845bfd806 : FAILURE in 55m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fc55205dcb06452eaf2ce83e43550482 : FAILURE in 1h 06m 53s","accounts_in_message":[],"_revision_number":18},{"id":"a010a565750c23fecee162a8fa37f1abe370e0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:55:19.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ec3c79380b74524b83c4addbd34ac77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ba06ad522e478b8fcb4182a235b1f2 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9342eed23cf40ba8362ee9364cb2840 : FAILURE in 8m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a99914ff7552469abd03bfe6f7cb6f31 : FAILURE in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad95cb3409094f16913732ceb8e898b9 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a5ed88a38e5748969b002d720470107a : SUCCESS in 28m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8e2e87781c04de885e23e9ffae42af3 : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9f305179aec48e5aaec9ee756c22f32 : SUCCESS in 1h 44m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/872955141e6041779ec65f2687c9ae73 : SUCCESS in 1h 03m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26e5b019db44478780986185e5b9c897 : FAILURE in 20m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/00460914142840569ce03c1ec088b61c : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3630319371944d65bb75097cab84f39b : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b11f34d602b24f8a97b20b24af0042ad : SUCCESS in 6m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/82068aba888040548db9d3df7d42a49d : SUCCESS in 8m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18f2c936e4b241f086721c102aa641a4 : FAILURE in 41m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0b0fe98fde34bf4889d1c3f01a3e739 : SUCCESS in 4m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e11ef442cd248a18e845b298a130d16 : SUCCESS in 1h 00m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d09b685b9874d9b9802910c42839068 : FAILURE in 1h 39m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f912f7e127ed4c50ad53060c238d7dd9 : SUCCESS in 1h 55m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/185b14d56bb948b2b7f96e8c35ed1d09 : SUCCESS in 1h 30m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1bb88fd0957f40cd9c5b1a5e278a87e8 : SUCCESS in 25m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9601980f0e04cdd881d300849bb8cf4 : SUCCESS in 1h 09m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48823ef96bab4a4f888a03ac8b40919a : SUCCESS in 1h 26m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76d196dc159246d29e2428f18e3a45bb : SUCCESS in 46m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8dfa28eb1b247ada469863027f0a095 : SUCCESS in 1h 36m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ac42f8b1920143e380ee4f64ee45f648 : SUCCESS in 1h 33m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d9cd2ed8a774243a2f70600ac369e64 : SUCCESS in 1h 30m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c41cbee3c3f74903ab24777f5d470283 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":18},{"id":"125eb212e38c611ee0412bec3f9f5eb47a9e8d33","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:16:41.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2500/.","accounts_in_message":[],"_revision_number":18},{"id":"e68137c32a92f4d8c9b6b8e14bd68aae307c281a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-11-19 18:02:19.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"78c310f832f8d85e0d3f6b959cd219ad7029dd84","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 18:02:46.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2519/.","accounts_in_message":[],"_revision_number":19},{"id":"04ba0ffef86460ee7bd106a5c87799c539a26987","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-19 18:11:15.000000000","message":"Patch Set 19:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/15/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"9949c9c2d0fcbfc890834b749bd8d14b0f9dc708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:31:33.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4393730e3040d88157875bf412bf67\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df94c471d4a74423bf6ea2e012d12612 : FAILURE in 25m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1457d7100f844bf0adac324e871b26e9 : FAILURE in 25m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da9bc6e436d443c48297c26f85021f23 : FAILURE in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"da802462af730153676f0b7207701f2a9381e4fb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-19 19:04:27.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":19},{"id":"db94968a874fc7f32b024da9246cd451cad2b91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 20:13:31.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4483b8ab239848fc8e74c30bbf22b1db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562cfb7169b04829802cbbbc43481e18 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14f89dd8b5a04bfdbfe5b6aa62851923 : FAILURE in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b8125977de641fe87518da726d6d2e9 : FAILURE in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07bf4a0b9f75478aac2d70125d6b8a48 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/4e87e17063de460f9ae503f92ae5321f : SUCCESS in 46m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1221d1370c744facade9f49a05d37ed3 : SUCCESS in 34m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b8f3bb8c3584ce1babb0d9cde727a96 : SUCCESS in 1h 24m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcab6e36b73141de92435f350004f156 : SUCCESS in 50m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d498ecb7e144e3eb82699f9f4d3111a : TIMED_OUT in 40m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5f38eb3c5537484e98cf2d1c91b5e77a : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/022cf2915f9a412f8ffdd34ac6f3c973 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a07a02db6baf40a188fa8218a919c816 : SUCCESS in 4m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5945201aff864076b3cd81f42837c957 : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1c2d29486a74042a257dcbf2213d070 : FAILURE in 21m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4881a0b1a4164df78ff5413e2dfb714e : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6df44e83641442feb9face44b24f5602 : SUCCESS in 1h 48m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e812ce5a0fd4dcc8c15121d9f27346d : FAILURE in 1h 55m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b024aaf830ad4b28a7cdf156958c9f41 : SUCCESS in 2h 05m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3e56c33074794bba87b5a104a710d57e : SUCCESS in 1h 51m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5cca3c7d28f142418fcc75b145b13e12 : SUCCESS in 38m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f4d2c7cf85c41cd8e191f65f297a8a7 : SUCCESS in 1h 14m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df1085ba9d9f4ab9ad0cdb04c89b1a73 : SUCCESS in 1h 40m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5c457ff1b8a4c7f880eb9774417435c : SUCCESS in 28m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8fab5d27abc483cb45ec9bfba25056a : SUCCESS in 1h 27m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/879007d61cd34b35a0388a46a95a131e : SUCCESS in 2h 06m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e9a0bb81d264ce6aec1ec381d63f0ee : SUCCESS in 1h 50m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15756219c02340c48d926d1497b11e18 : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":19},{"id":"87f31678dc6fc0f1b08dd605a30346e46a360beb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 01:29:41.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f4f063479d284471a6ca67d6fbbc8015 : SUCCESS in 46m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d65506c50b1e418e9d2c5208315508a9 : FAILURE in 48m 12s","accounts_in_message":[],"_revision_number":19},{"id":"a6c639bce4f3c63cd234eddb421a2c7b20182eed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 03:24:39.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 10m 27s","accounts_in_message":[],"_revision_number":19},{"id":"d8826dc882b2163f8a01c0b459a7f131e3fa3835","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 08:51:02.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2663/.","accounts_in_message":[],"_revision_number":19},{"id":"d16ed60f69377b989101e71591028013672fbca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-11-20 22:56:28.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"36144f58f0755cb8168f6f54593956ed86b4f918","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 22:56:52.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2700/.","accounts_in_message":[],"_revision_number":20},{"id":"fe5eaff05de8fdd7699eddb3c3ebe7ecf8854e00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 22:56:54.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2557/.","accounts_in_message":[],"_revision_number":20},{"id":"257cab661755134f10240f063117b736421beacb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 23:16:06.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/51/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"ddb2f6d25f5eb398ac07698e35ab2e1e3c06bdc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 23:38:38.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6913e8f6a5e4127b5428f03f9e60ae2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9de41ed9236543c3a83ff794c988f303 : FAILURE in 13m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64eddb1212c641b390b096fbeddee64c : FAILURE in 16m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05349c9a48964449be6dd47a1151f853 : FAILURE in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"42675948e212d891504ea061c143cb893f3c1939","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 23:52:04.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":20},{"id":"e50c16d698de5dff38fe1896b5974e1e7b9246a5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-21 00:09:30.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 11m 36s","accounts_in_message":[],"_revision_number":20},{"id":"eae2e19bf9dad305c90139e1fa6d21cd4a20bfc8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-21 00:36:58.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7b454e4e2a2942ae9626a9a32411b993 : SUCCESS in 47m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f7c0f5f8d76542e7a6055586a3b44d8e : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":20},{"id":"604ba50f43c424d334dc21113416420dd1feaffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 01:33:24.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/515dc47e0c59499884adc67a5c5203a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1acf1f830dc64b2baf1344687705caec : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf86083a7f4b41a69e8fa67060e11d59 : FAILURE in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07b8a3ceaa4649299f4ac2ee2b9dc730 : FAILURE in 14m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e734d479c4d74986a2658767bbb32c45 : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7737e492660340e79fcf6cce94cb6efb : SUCCESS in 33m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6477aaeb89a4d91b493f3281609d384 : SUCCESS in 53m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfb59a28742c4c41a61b7ccc67ebb6b0 : SUCCESS in 1h 35m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/424faabd318a4375ae0e0bcff48aa4c4 : SUCCESS in 1h 04m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90c0c255b86e4d49932838f417cb0b1a : FAILURE in 18m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a24cab5089044269915a56efa8a933f6 : SUCCESS in 3m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04c23f17299c413e82e10f676ff7395a : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c907c1ddf11e4f9f8df1de2d43e34226 : SUCCESS in 6m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/507cd5bf4da7458f9a5ad83229b0730a : SUCCESS in 9m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0cd6696f821140e299bfb962d719c87b : FAILURE in 19m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/301acec2412c4db2937784bd4a9ac4bc : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2236e663d3b40d68b8edd3fda976d5b : SUCCESS in 2h 00m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad3a34bdc57944498808165694113717 : FAILURE in 2h 22m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80b6e77f528c43efafd65ddd888b9b81 : SUCCESS in 1h 55m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d9b10d324fab448abd0fc8c0146757c8 : SUCCESS in 1h 56m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1673e4a380f949fd9057e176f8974211 : SUCCESS in 23m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e256944137244bebbdf119c1dc95b12 : SUCCESS in 1h 10m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/41fc895973eb40c7af877f6753169fa0 : SUCCESS in 1h 57m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/09e27cb3f6f140499704d7e0068feff0 : SUCCESS in 26m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa92ab7345c54b148701365fd94f052b : SUCCESS in 1h 30m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/25fe814cc8d94c5ea5a906cfd871cdc2 : SUCCESS in 2h 31m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f972251ad13840c39085d80134d29209 : SUCCESS in 1h 58m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45bb63aaf5f04e34a0d8fcd7075dbc2a : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":20},{"id":"6cba3f8c560518dcc23bd0deb7cad63c7f2b087b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-12-02 18:44:08.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"cb94616b22b9ed09f6e4fa3f30612209b1995543","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-02 18:44:35.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2762/.","accounts_in_message":[],"_revision_number":21},{"id":"19dbaa8b168b0bb34e07a119d32424cb5d829c9c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-02 18:44:35.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2614/.","accounts_in_message":[],"_revision_number":21},{"id":"4e905d482ecfaaab1ea2868577c17b2d968a7add","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-02 19:03:26.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/112/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"d647b7592a64701bb365c8b5cb14e26519509761","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-02 19:41:56.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":21},{"id":"24d71dc3edea3504a1f3817ba93329b81eb4861c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-02 19:53:04.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 07m 35s","accounts_in_message":[],"_revision_number":21},{"id":"9b82f32037f55356b19c982fae49d9c57338e36c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-02 20:54:16.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8ab87aae2cd8464ea5f09ea2cef63528 : SUCCESS in 54m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/135a6703b3d943ce84070fd078b1881a : RETRY_LIMIT in 2h 06m 23s","accounts_in_message":[],"_revision_number":21},{"id":"361b2512a5747bdd0ead5af13a2b962e9b52f84b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 20:59:01.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75fd220e66824b509be56b188e408fca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961c7d70b558443c9995469f084214b2 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/706ab4b619334ed992b0b01d58b2a8cf : FAILURE in 13m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a37617e18b1342d0b55c65e82d4a3dc8 : FAILURE in 14m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc287ea1ef204d8a9beff699b01e1231 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a2dfbef4172242f0956d187befb278a0 : SUCCESS in 45m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/442630f037df4ca7a96087c479d56313 : SUCCESS in 46m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87f87d5af11c4a689469f4d86056d690 : SUCCESS in 1h 37m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ab17e442ec04e7296a303f2dc7545fb : SUCCESS in 48m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dabc94f87bcf4599804eaca4063fb08d : FAILURE in 24m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/96b2d6bef84743849b7bb67310ce4008 : SUCCESS in 6m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/439dc7f3dff240a78d06ffd9e515dfa0 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf313d43ae4c4ac590e0d6f4f362c6fa : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d79599fd23d44c48bd119f5c62f9305b : SUCCESS in 8m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b35a7c49183b4ecf9c1832f4acf4473b : FAILURE in 32m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f1acf6c9c0245a1951a495e68580031 : SUCCESS in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6176a8efea74c1ea098fb1c33f1b587 : SUCCESS in 1h 38m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1af20bfc99b74582b4448b19cb0cb20f : FAILURE in 2h 06m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dca8717f19394ceb97da2da9da0a4120 : SUCCESS in 1h 36m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/363a1ddcc41e408983346413ad50268b : SUCCESS in 1h 46m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/27336d29a88e4795967420fca6bb7759 : SUCCESS in 32m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/162dd870be9941c093ff9e8b6d50c82b : SUCCESS in 1h 03m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/079c54c21a5c4d10934424d7c2508cda : SUCCESS in 1h 21m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38c9b41b46c94f7ea03333ba1b8fa504 : SUCCESS in 49m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b12fa8265d4d4414a87f033e5b1eb9c2 : SUCCESS in 1h 45m 49s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/25483214acdb474ea75da64362318b59 : SUCCESS in 1h 50m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/283b3219294e422fa542331f45c7563e : SUCCESS in 1h 40m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e6d09b022a143c8bb38bb0342d3f99f : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":21},{"id":"21b5ada22fad35419ea1c6c8ab9bc56b927fd3c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 00:44:39.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a739c50462a441e9aa7ce73655e1eede\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2deda8dceb8646cdac56ce3fcf54ccc9 : FAILURE in 14m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/997dfdc7a75e42e1b37a3e0e7b46c2be : FAILURE in 11m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08938d457be84e759d8bbc4075f6068f : FAILURE in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"82e5191515d91b2c4a5b867e1608768b817d2d13","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-12 06:22:41.000000000","message":"Patch Set 21:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/21/20251212105053/NEC-V-ISCSI : FAILURE in 3h 37m 41s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/21/20251212105053/NEC-V-FC : FAILURE in 4h 31m 45s","accounts_in_message":[],"_revision_number":21},{"id":"922ad5883ee1a93ddabd80da86529b1cabb25f0b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-12 09:58:08.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-21-fc : SUCCESS in 1h 5m 48s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-21-iscsi : SUCCESS in 1h 5m 48s","accounts_in_message":[],"_revision_number":21},{"id":"2f46bfd5accf33c8803e2cb9b91a617e5fc40327","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-12-12 22:10:09.000000000","message":"Patch Set 21:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/21/20251212210121/NEC-ISCSI : FAILURE in 9h 44m 26s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/54/939154/21/20251212210121/NEC-FC : FAILURE in 10h 08m 46s","accounts_in_message":[],"_revision_number":21},{"id":"7a7c8d8f256b02141a1744159e9fcedf9a488c46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-12-22 19:13:45.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"88b8c6cb1da8f950e91e186c1426bee5cc49df99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 19:44:25.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313eda1c79a44950874311b97595f368\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dad16e5fd094444182162e98c1636608 : FAILURE in 24m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de0b9533da6b44d5b6fdbc80744ba7f4 : FAILURE in 23m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/191d04616a50406d8f64091d989e92f6 : FAILURE in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"67384c51b96bb140022baea7cf0cfb80360b1ea6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-22 20:05:23.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9a467f4b553f4bf2840912474dc7ead7 : FAILURE in 44m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/93b3cde60e774180af132e8e37202c26 : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":22},{"id":"f619cfee82eed29a42dc47d529bb3a65301021a0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-22 20:08:17.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":22},{"id":"45b5fce8450a051027bf881c13d20f0aba129863","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-22 20:12:35.000000000","message":"Patch Set 22:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2957/.","accounts_in_message":[],"_revision_number":22},{"id":"ea835c9da1fab8bf0bed6b3f2b448a42a54cee16","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-22 20:13:49.000000000","message":"Patch Set 22:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2786/.","accounts_in_message":[],"_revision_number":22},{"id":"e41993565159a2f254923465121b2402dc5dcf55","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-22 20:34:17.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 19m 02s","accounts_in_message":[],"_revision_number":22},{"id":"ceec2112961345c22722f04e3a23b6d7fd603e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 22:29:10.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbaec55a5df549edba38852027932113\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20468199e0874f3daf3fc6ae2a81718b : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f41b03dc8614ec9a0515489107d78ed : FAILURE in 10m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a09d465ca7804aafb11b60de2fbaae10 : FAILURE in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31222ffe5244ea2962a7d8e480c0498 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/28d4ee64398f4b2586505b2eb422dab1 : SUCCESS in 29m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b306cbadf6124f5ebd2d320a5a0cac34 : SUCCESS in 43m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdedaa65412a49809f914cdc944e5b64 : SUCCESS in 1h 28m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bb815869ce445e5a59d5c27887703d3 : SUCCESS in 1h 01m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5250b73c996c4f969630d2bf7b8f4ab8 : FAILURE in 25m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/29dbe996b447475cbad0ccaf99da4b29 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/536c30ae47784371b4d75be9db86ac57 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1bfc6a91c1fe4745b30052c76ed2047b : SUCCESS in 3m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9bb8cce2ae5d4836bf281132ffa1a457 : SUCCESS in 7m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/adcd465c71e14228b56b694e9b3f7983 : FAILURE in 20m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8718a45865304f7aa859ff4a5c3af3f2 : SUCCESS in 3m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/358af86a13ad4496852f0db7cd416c63 : SUCCESS in 2h 12m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/13209db886b44306b2f3a8308403f92b : FAILURE in 1h 29m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/31cb1ae3928f4bcdbb33126906fdc198 : SUCCESS in 2h 15m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e329919b6d4c4fadbf9eedd3dc58c487 : SUCCESS in 2h 21m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/21c64411fd41492397f7034c03cdeee3 : SUCCESS in 20m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f0a9003661f4222bdda160b4cbbdaae : SUCCESS in 35m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36ab9de7bc594bbe899e64c8bf8b2951 : SUCCESS in 1h 08m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2b5dc5269f2741c8adc0a06283965b05 : SUCCESS in 31m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/af301c293ff442d78ce39f2dd1cfe601 : SUCCESS in 2h 04m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bcb8992726584a56b1c8b48d70d5629a : TIMED_OUT in 3h 10m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40f3df9edd7d42648f1467868b4364af : SUCCESS in 1h 38m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0bd782685148279742a719e2893608 : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":22},{"id":"b1b84c36e87190d5781dd953951000f24d082183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-12-24 18:20:17.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"17fa36948cf1b82525a845b0f980855c12a6b889","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 18:53:56.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337de2d32db24d9ba0c6deafe21658cb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b4d3f07bbec49ec998533483fdd6b56 : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/471b3492b9bf403684097c7073ef5ad3 : NODE_FAILURE Node(set) request 48cd5e94b054405e8f92e1ded5f7a2d4 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0673e9b02f8541809e7592aaa5d08303 : NODE_FAILURE Node(set) request 26ef68724a324ff4a7751fc9add49fd8 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"98d99dbf30c687de77ae3e19c0d798505662bdcd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-24 19:13:45.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0cd8841b0fa8406b93a7a0abdc38b8f5 : SUCCESS in 49m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/80787d34b77c4b8796d659958315a663 : FAILURE in 49m 17s","accounts_in_message":[],"_revision_number":23},{"id":"b337310b1f708b896b59ffeb8122a881d0bcdbcf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-24 19:20:59.000000000","message":"Patch Set 23:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2804/.","accounts_in_message":[],"_revision_number":23},{"id":"fc71cac96fffadf2b99ffc0d99d8d4ac25b8d673","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-24 19:21:56.000000000","message":"Patch Set 23:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2975/.","accounts_in_message":[],"_revision_number":23},{"id":"8cf357fd5775ce82d2df9ee029403003b43f744a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-24 19:24:00.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":23},{"id":"306ffce2af1e326679c3b0196e6f91fbad1eaa2d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-24 19:40:42.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 19m 05s","accounts_in_message":[],"_revision_number":23},{"id":"21271fde33fc94a0a9cb6baed7fcc5e9ba7da768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 21:24:06.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658dc71d35754ce6977a7478455f6244\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9b120abcaa4d45bb16ef8acadf9d2e : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78d54befcf3e49b4b41ebb928dd634ee : SUCCESS in 8m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d88e93e19244ea0b66d5bc4b3fedf76 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30099822357d4dd9a36aa25f145614b2 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/623cb4933baf4b3da8caf6735b33007e : SUCCESS in 40m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9f88ebeb02d4c159e68598f2c3e3e7b : SUCCESS in 31m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fa748bcc4834251837ce384f1e75d4c : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af402e147367484a9f2369c167664dd8 : SUCCESS in 32m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/165d40198f034dcf970d307f5054f7b6 : SUCCESS in 16m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b01f08c266bd46118a05b2f6b3c5d260 : SUCCESS in 6m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a0cca93a43d45a88e7182b679d3756b : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff3c4715bfc045f7a154017d2c480d06 : SUCCESS in 3m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2be16432fdb34ffea96a285c7249a719 : SUCCESS in 9m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/874315dcc8584b08b124190cd630e222 : FAILURE in 21m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6abd5bc4b8ff4913b9f11e3dec13c534 : SUCCESS in 3m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a5ab55e403d54ab9ac7a1f1ce32c61f7 : SUCCESS in 2h 02m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c42de9c7024944999b10d4c154a5726f : FAILURE in 2h 58m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55d0bb209080456a9229425b190f9644 : SUCCESS in 2h 01m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df33b1a4f603496ebe5d58cc61713d44 : SUCCESS in 1h 46m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f6e0e6bfbc4c49f4b7808efd24208c07 : SUCCESS in 35m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed8e2df472204437bdddd4b1625a6031 : SUCCESS in 35m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7367925f8544bfca612b0b41b056e86 : SUCCESS in 1h 34m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0238d719406147caa3f9c0a2e168c54f : SUCCESS in 38m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8a4efbe6121e4180baabf4c5ec86e4d9 : SUCCESS in 1h 11m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5129832f4d99441996cfde51cdbd9fe1 : POST_FAILURE in 2h 21m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/153a2a2fee5f4cbdba9633dcdae8077a : SUCCESS in 1h 50m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f50e292bd0c423a820f523aa9a516cf : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":23},{"id":"dfefd36e023140f657dab4ad44a2876726d7a3f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-24 21:31:13.000000000","message":"Patch Set 23:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_12_2025_21_24_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":23},{"id":"f17b608df0f93750166af56072eddaa1dea3cbc2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-25 00:02:22.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-23-fc : SUCCESS in 59m 3s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-23-iscsi : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":23},{"id":"4475cc20a6b56dacf5ecc589f035ba7c96a99bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2026-01-05 21:36:35.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"7f3af0e4c77c28b5280165bdfabfa8624dc1b8fb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-05 21:37:03.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3ae22c9623d64202811edeb0a080a857 : NODE_FAILURE Node request 200-0000942324 failed in 0s","accounts_in_message":[],"_revision_number":24},{"id":"45a7ad7107129bd7e9c3f6e00f150c1553219684","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-05 21:38:37.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6c01cbeab3874d1bb42ffc42f7ddf945 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7d0df6e8fa9e488c81bb32435b41f301 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":24},{"id":"846ddabc1a03cad9fca3292e8f25c46ce736ad17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 22:13:15.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ab574c1d5b40988c0b5552640a3eb6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0cde4b9ecec140618e232218ddf270d2 : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbda0356802940ec86cbd2f4dfbdba14 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41637a2b24cd4f2faee95f7082dff7b0 : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"50af41cafc496184c9ae13cc97c3bee351304705","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-05 22:33:24.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I5caf27edfb690e37a91002285aca88e870b4c77b .","accounts_in_message":[],"_revision_number":24},{"id":"96b3dbe6457c68d2859870c4b5261b0ae76a7b24","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-05 22:48:22.000000000","message":"Patch Set 24:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_77/index.html : FAILURE in 71m","accounts_in_message":[],"_revision_number":24},{"id":"e5333cd02beaf3233fdcc779457994ea4d501d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 00:07:46.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5443e528d1a4db280147aa473347dc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61719400aac4276b8d2cc5d9c56e9bf : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0524cd1232ef45a2863db8d5d7018941 : SUCCESS in 10m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/578f5b27ad13441f9f91f23e14b1c936 : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab08df4da31b42578e4d686281b1beea : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9307a65ecb5c47709b4f440add5f997f : SUCCESS in 53m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3da1c55b6189452191bf3294f3380a4f : SUCCESS in 56m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70fb8feb15474503888481e7800f5f2d : SUCCESS in 1h 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa670ee64202423b80cf7b658023649f : SUCCESS in 59m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8723f7d83f24fc3898b7aaf9a984606 : SUCCESS in 18m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4ad3bf32886e4077aacd64fb0de35ab8 : SUCCESS in 3m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a9a8f3f08ee459287d01f05a236784a : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3287fe3e78ea4513a72bcfb8623a4c1c : SUCCESS in 5m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1086cdd544224024a80a1692a9cc206b : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/da93bb93fd9d4ca99db4cee75e2d7b4f : FAILURE in 36m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29e759ef2a554d54b035c2d73ff359c7 : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a92312ee5fbb40f498e4ca47eaec5d83 : SUCCESS in 2h 20m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f0583cfa20f34c668bfa0690099f0165 : FAILURE in 1h 41m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd32d4b42ac648008e62209fcbc60a24 : SUCCESS in 2h 08m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/17641aa1ba7b4c70b9ddc59081f7e022 : SUCCESS in 2h 10m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b178021a5c14f03bf3f415a0f6f346f : SUCCESS in 23m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6568628dddbe46d59242e70eb756d275 : SUCCESS in 1h 04m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ed019c97a9745f8b261437cf16091cd : SUCCESS in 1h 42m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a01494fb0cfe4e598473e3186e016ed7 : SUCCESS in 43m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b093e6a44a5c4e76be8612232d942e5c : SUCCESS in 1h 59m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e4773f8663de4d58a28949dd4ec02384 : SUCCESS in 2h 24m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fb4ccae9e9c4623b91a72bb4179b73b : SUCCESS in 1h 36m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ac8785c0144f54b002a3ee88443906 : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":24},{"id":"9251b426b37b6249e21cfe17a31523469607c835","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-06 00:14:35.000000000","message":"Patch Set 24:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_01_2026_00_07_56\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":24},{"id":"a57261b4f90038b567b5fc25c78dd3fd41c09265","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-01-06 00:17:16.000000000","message":"Patch Set 24:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-24/FC/ : SUCCESS in 2h 40m 18s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-939154-24/iSCSI/ : SUCCESS in 2h 23m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":24},{"id":"0946565bdbea1f8b0aba7933c3fb03d2f644d6cd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-06 01:50:25.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939154 : FAILURE in 1h 22m 23s","accounts_in_message":[],"_revision_number":24},{"id":"5e503050c608a15e361a7abc05298106f0648d60","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-06 02:52:20.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-24-fc : SUCCESS in 54m 59s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-54-939154-24-iscsi : FAILURE in 1h 17m 4s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"26231f4a5e6156b1ac9e61a0d8bdf6e4a4170676","revisions":{"53c8819cbefb5330d3b69a353ece3518bbf70f6f":{"kind":"REWORK","_number":1,"created":"2025-01-13 18:54:36.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/1"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the project UUID\nin the URL and project UUID in the request context to check if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53c8819cbefb5330d3b69a353ece3518bbf70f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53c8819cbefb5330d3b69a353ece3518bbf70f6f"}]},"branch":"refs/heads/master"},"c5005e9cfcc62c17096b5f3b4151ceee3f7c797d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-13 18:56:00.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/2"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:55:25.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the project UUID in the URL\nand project UUID in the request context to check if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5005e9cfcc62c17096b5f3b4151ceee3f7c797d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5005e9cfcc62c17096b5f3b4151ceee3f7c797d"}]},"branch":"refs/heads/master"},"e7811172d374a01964de2d5355c0dc8a5ca65b47":{"kind":"REWORK","_number":3,"created":"2025-01-13 19:02:16.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/3"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 19:01:46.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7811172d374a01964de2d5355c0dc8a5ca65b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7811172d374a01964de2d5355c0dc8a5ca65b47"}]},"branch":"refs/heads/master"},"6d8d6c9eb3899964048ac2298f5309a58e056301":{"kind":"REWORK","_number":4,"created":"2025-01-13 19:21:15.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/4"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 19:21:03.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d8d6c9eb3899964048ac2298f5309a58e056301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d8d6c9eb3899964048ac2298f5309a58e056301"}]},"branch":"refs/heads/master"},"811b55115380b72bf486d8cc1a364204add227fe":{"kind":"REWORK","_number":5,"created":"2025-01-23 20:02:27.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/5"}}},"commit":{"parents":[{"commit":"962fe29e778c58a8e90c78602e7249cb4b06e450","subject":"Merge \"HPE 3par - Fix retype volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/962fe29e778c58a8e90c78602e7249cb4b06e450"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-23 20:02:12.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/811b55115380b72bf486d8cc1a364204add227fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/811b55115380b72bf486d8cc1a364204add227fe"}]},"branch":"refs/heads/master"},"28ed0c735c5030bf0daac4b260e100650d889c0f":{"kind":"REWORK","_number":6,"created":"2025-02-10 22:14:48.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/6"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-02-10 22:14:37.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28ed0c735c5030bf0daac4b260e100650d889c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28ed0c735c5030bf0daac4b260e100650d889c0f"}]},"branch":"refs/heads/master"},"b72e1be5331b2eb2f39efa45a7b4463597a4b122":{"kind":"REWORK","_number":7,"created":"2025-02-11 20:06:37.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/7"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-02-11 20:06:25.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b72e1be5331b2eb2f39efa45a7b4463597a4b122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b72e1be5331b2eb2f39efa45a7b4463597a4b122"}]},"branch":"refs/heads/master"},"e9a4478617ccd07ee684ab21b27689bbb78e9671":{"kind":"REWORK","_number":8,"created":"2025-03-12 21:51:56.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/8"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-03-12 21:51:50.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9a4478617ccd07ee684ab21b27689bbb78e9671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9a4478617ccd07ee684ab21b27689bbb78e9671"}]},"branch":"refs/heads/master"},"df17a84117a784adaf15ec55ef6723dbdc57dad4":{"kind":"REWORK","_number":9,"created":"2025-03-18 17:59:53.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/9"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-03-18 17:59:45.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df17a84117a784adaf15ec55ef6723dbdc57dad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df17a84117a784adaf15ec55ef6723dbdc57dad4"}]},"branch":"refs/heads/master"},"3f0a86cb068cc68ad6e7332f6d874e9428b5301a":{"kind":"REWORK","_number":10,"created":"2025-03-20 17:09:45.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/10"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-03-20 17:09:41.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f0a86cb068cc68ad6e7332f6d874e9428b5301a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f0a86cb068cc68ad6e7332f6d874e9428b5301a"}]},"branch":"refs/heads/master"},"2209b873282e72eb8e7e21eef25282a8356e3da9":{"kind":"REWORK","_number":11,"created":"2025-03-27 18:34:07.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/11"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-03-27 18:34:08.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2209b873282e72eb8e7e21eef25282a8356e3da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2209b873282e72eb8e7e21eef25282a8356e3da9"}]},"branch":"refs/heads/master"},"1d9b8319a96921fcfa8a573b50a02a5257c1f2d0":{"kind":"REWORK","_number":12,"created":"2025-04-01 17:11:48.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/12"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-04-01 17:11:53.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9b8319a96921fcfa8a573b50a02a5257c1f2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d9b8319a96921fcfa8a573b50a02a5257c1f2d0"}]},"branch":"refs/heads/master"},"6ba32c61129408e8ed9949f56a89ba022ece4b75":{"kind":"REWORK","_number":13,"created":"2025-04-01 17:12:34.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/13"}}},"commit":{"parents":[{"commit":"a1fb7da213b72040b8c3fa2b2708f23c21da20b2","subject":"Merge \"HPE 3par: Update the calculation of free_capacity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1fb7da213b72040b8c3fa2b2708f23c21da20b2"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-04-01 17:12:37.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ba32c61129408e8ed9949f56a89ba022ece4b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ba32c61129408e8ed9949f56a89ba022ece4b75"}]},"branch":"refs/heads/master"},"5e707e0f85539581791bc7249a29460ff6199fd4":{"kind":"REWORK","_number":14,"created":"2025-04-22 18:41:07.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/14"}}},"commit":{"parents":[{"commit":"a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9","subject":"Merge \"NFS driver: Fix fail creating volume with multiple snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-04-22 18:39:54.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e707e0f85539581791bc7249a29460ff6199fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e707e0f85539581791bc7249a29460ff6199fd4"}]},"branch":"refs/heads/master"},"6684ac71c54a81006b84f96d60a9fa3252981fe1":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-06-02 21:06:10.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/15"}}},"commit":{"parents":[{"commit":"ae78fd97b1b1c4ffb2eb8042801a0d5fbef8eff4","subject":"Merge \"NetApp NAS: Update NFS online volume extending support matrix update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae78fd97b1b1c4ffb2eb8042801a0d5fbef8eff4"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-06-02 21:05:56.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6684ac71c54a81006b84f96d60a9fa3252981fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6684ac71c54a81006b84f96d60a9fa3252981fe1"}]},"branch":"refs/heads/master"},"04d3940a1bb80f4821658cd68dd3effe8aeac63f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":16,"created":"2025-07-08 20:54:02.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/16"}}},"commit":{"parents":[{"commit":"3380f01ff2876a6fdb6d4fe94984936277aefe90","subject":"Merge \"Hitachi: Fix to have compatibility of GAD volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3380f01ff2876a6fdb6d4fe94984936277aefe90"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-07-08 20:53:32.000000000","tz":-420},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04d3940a1bb80f4821658cd68dd3effe8aeac63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04d3940a1bb80f4821658cd68dd3effe8aeac63f"}]},"branch":"refs/heads/master"},"38ef22dcf578b90d36b9d2bb184ae209a5f37be4":{"kind":"REWORK","_number":17,"created":"2025-11-13 21:22:00.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/17"}}},"commit":{"parents":[{"commit":"2c90afcbaa65bb298ee58fc325f4e5d90a4eda94","subject":"Merge \"docs: update data leakage docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c90afcbaa65bb298ee58fc325f4e5d90a4eda94"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-11-13 21:21:49.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ef22dcf578b90d36b9d2bb184ae209a5f37be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ef22dcf578b90d36b9d2bb184ae209a5f37be4"}]},"branch":"refs/heads/master"},"fee84a67aeed98a6a4eb6c320597076096e84d36":{"kind":"REWORK","_number":18,"created":"2025-11-17 18:57:55.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/18"}}},"commit":{"parents":[{"commit":"2c90afcbaa65bb298ee58fc325f4e5d90a4eda94","subject":"Merge \"docs: update data leakage docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c90afcbaa65bb298ee58fc325f4e5d90a4eda94"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-11-17 18:57:42.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fee84a67aeed98a6a4eb6c320597076096e84d36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fee84a67aeed98a6a4eb6c320597076096e84d36"}]},"branch":"refs/heads/master"},"04b2d30f00ac6353f47d0022244f4e550c4fb2d6":{"kind":"REWORK","_number":19,"created":"2025-11-19 18:02:19.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/19"}}},"commit":{"parents":[{"commit":"2c90afcbaa65bb298ee58fc325f4e5d90a4eda94","subject":"Merge \"docs: update data leakage docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c90afcbaa65bb298ee58fc325f4e5d90a4eda94"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-11-19 18:02:02.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b2d30f00ac6353f47d0022244f4e550c4fb2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b2d30f00ac6353f47d0022244f4e550c4fb2d6"}]},"branch":"refs/heads/master"},"df93a21f3c09ada49f3ad3fc8367887ba69564ac":{"kind":"REWORK","_number":20,"created":"2025-11-20 22:56:28.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/20"}}},"commit":{"parents":[{"commit":"2c90afcbaa65bb298ee58fc325f4e5d90a4eda94","subject":"Merge \"docs: update data leakage docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c90afcbaa65bb298ee58fc325f4e5d90a4eda94"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-11-20 22:56:17.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df93a21f3c09ada49f3ad3fc8367887ba69564ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df93a21f3c09ada49f3ad3fc8367887ba69564ac"}]},"branch":"refs/heads/master"},"3136623b0f77437b1860b6ea53298d9e4546029f":{"kind":"REWORK","_number":21,"created":"2025-12-02 18:44:08.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/21"}}},"commit":{"parents":[{"commit":"0adfed073344468ceaec4d66cd9d4c0a51814e0d","subject":"Merge \"NetApp: Fix terminate_connection on unmapped vol\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0adfed073344468ceaec4d66cd9d4c0a51814e0d"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-12-02 18:43:57.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3136623b0f77437b1860b6ea53298d9e4546029f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3136623b0f77437b1860b6ea53298d9e4546029f"}]},"branch":"refs/heads/master"},"e06453150eef6a055b5bec49945d98152ef9d034":{"kind":"REWORK","_number":22,"created":"2025-12-22 19:13:45.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/22"}}},"commit":{"parents":[{"commit":"2bdc892a023d2bbdf8e66e0e9645a86b776fcf61","subject":"Merge \"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bdc892a023d2bbdf8e66e0e9645a86b776fcf61"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-12-22 19:13:35.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e06453150eef6a055b5bec49945d98152ef9d034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e06453150eef6a055b5bec49945d98152ef9d034"}]},"branch":"refs/heads/master"},"38e1027f0e85dd3cbee373795c617ab9c7f9d772":{"kind":"REWORK","_number":23,"created":"2025-12-24 18:20:17.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/23"}}},"commit":{"parents":[{"commit":"2bdc892a023d2bbdf8e66e0e9645a86b776fcf61","subject":"Merge \"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bdc892a023d2bbdf8e66e0e9645a86b776fcf61"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-12-24 18:20:06.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38e1027f0e85dd3cbee373795c617ab9c7f9d772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38e1027f0e85dd3cbee373795c617ab9c7f9d772"}]},"branch":"refs/heads/master"},"26231f4a5e6156b1ac9e61a0d8bdf6e4a4170676":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2026-01-05 21:36:35.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/54/939154/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/939154/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/939154/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/939154/24"}}},"commit":{"parents":[{"commit":"00505d25879992b268402a71125ccc65b3c4088d","subject":"Merge \"mypy: NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00505d25879992b268402a71125ccc65b3c4088d"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-01-13 18:52:13.000000000","tz":-480},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2026-01-05 21:36:23.000000000","tz":-480},"subject":"_process_stack in wsgi disables admin API policy","message":"_process_stack in wsgi disables admin API policy\n\nThe _process_stack in cinder/api/openstack/wsgi.py will handle the\nproject UUID in the URL and project UUID in the request context to\ncheck if they match.\n\nIf not, the wsgi will raise the 400 error.\nBut this behaviour disables the admin API policy in Cinder, which means\nthe admin API with system scope can\u0027t operate resource in other project.\n\nCloses-bug: #2094009\nChange-Id: I5caf27edfb690e37a91002285aca88e870b4c77b\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26231f4a5e6156b1ac9e61a0d8bdf6e4a4170676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26231f4a5e6156b1ac9e61a0d8bdf6e4a4170676"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
