)]}'
{"id":"openstack%2Fcinder~939350","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I3c638dab2f0b5a2128eaeeea533450d7cdfa1577","project":"openstack/cinder","branch":"stable/2023.2","topic":"bug/1989514","attention_set":{},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-01-31 05:04:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c638dab2f0b5a2128eaeeea533450d7cdfa1577","subject":"NFS update volume attachment format during volume snapshot","status":"MERGED","created":"2025-01-15 14:14:58.000000000","updated":"2025-01-31 05:05:13.000000000","submitted":"2025-01-31 05:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939350-bug/1989514","meta_rev_id":"aead13c714b611f50e61959fe8215d32be8308d2","_number":939350,"virtual_id_number":939350,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-31 05:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-01-29 23:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-01-30 14:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-01-30 14:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 14:14:58.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2025-01-15 14:25:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-15 14:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 14:43:19.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-01-15 14:51:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-01-15 15:21:28.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-01-15 15:24:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 15:29:44.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-15 16:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 16:21:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-01-15 17:43:47.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 19:21:59.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-29 23:01:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-01-30 14:07:58.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"ddb7ec0559d607407cc5c6fe0cd072e5b1de19e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 14:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd8a39b74c232c23b1af248fb54ca4343f101ce6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-15 14:25:43.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/927de9bd0d69463aad476f8d3a8d8134\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d57f0962d61f4f7bb1210813887a659f : FAILURE in 6m 51s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61f4c134645e4ae691f88a4c4a49c5d1 : FAILURE in 6m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0edef1cef5f14cbc86e6a7f8e47ea1c3 : FAILURE in 6m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88cfc820bc4b4ee3a3c9d2300f394e45 : FAILURE in 6m 46s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003c... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"ba6aaf49a8f5ea5cf6f36b3f9f189fa88c3d548d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 14:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac25b4f4154c481bab03537040a661e7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12ed0901a4e74372b9276aafc24d668b : FAILURE in 20m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/231188e401914b49a9134dce7381ff5e : FAILURE in 19m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a47bc85dc2814ccbb61b10d079de6978 : FAILURE in 21m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0434a2be5ab84ba19c3bf264299621ce : SUCCESS in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47edda9c4b1f83bcfffe458e2ab2c3741c6ad54c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 14:43:19.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1219/","accounts_in_message":[],"_revision_number":1},{"id":"461d1c5a29341caa3a0df69310bd676bb02ca2d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 14:50:45.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939350/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939350/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"ae88be0fd9b25e0136c3de43e10bdf292ff56d95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 14:51:46.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/242/.","accounts_in_message":[],"_revision_number":1},{"id":"9174cf98ed3d41c0ad4ad8423f0afe44e6984997","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 15:03:10.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/230/.","accounts_in_message":[],"_revision_number":1},{"id":"0e964ebe83ded3bb73fb48d67776e85a428f03f6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-15 15:21:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939350 : FAILURE in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f1acb9d8feafb2b5ccbb375df76600a2e40db137","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-15 15:24:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939350 : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"697c0eb9c1a15d0879873dcf418018111b9185e3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-15 15:29:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939350 : FAILURE in 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"00b2afabe08972e9569e33b6fee6c2c6c2c1ced2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 16:15:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78c084f882a45d48aca81f968e53e1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a541789afe4b64bad2b3a24a8b0292 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34f60e67cb834686838531a5a533c471 : SUCCESS in 11m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a25a42d2bd7047bfb5c889e862a1f824 : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb18b0c607884dff8b3604a763effc53 : SUCCESS in 12m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d8a85fc358d42d4831beff9477899e5 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac3fdb8fd0d404587b4ab0cfccabdb5 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/93cf7badeaf74e51be8979d6e85c552c : SUCCESS in 53m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4cf14fce552c4b10a10fef6c37627880 : SUCCESS in 1h 01m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28cdc45df794420d9402f1b144e1ed21 : SUCCESS in 1h 02m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2098322f990414ab9dcfdddef53dfa1 : SUCCESS in 17m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c5d897975e114879bff7dc375874c92d : SUCCESS in 11m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/982a30cf5dbe49a5ab96a9d326330913 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4a91365334f4cf1ba3e3e22c5ad1dc2 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/591a027190ba481f80bdbe2ed4a99313 : SUCCESS in 10m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d56f876435d24e3c8c127cba5849c5fa : SUCCESS in 7m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f5af8aa8189646fb8c704efaf31b79ba : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4228ee88e3c42638c984c3baf145835 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee531168468c4f1eb231122794bee8e5 : SUCCESS in 1h 52m 00s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c1af7e2307342e7a2304d3dcd2dd629 : SUCCESS in 1h 47m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/51501d4a19dd4850b047bb2317673b60 : SUCCESS in 1h 41m 31s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9be2e6f74bae4c729002255ebb130ec6 : SUCCESS in 1h 12m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94e701b37956470dba8585d5216b8fc8 : SUCCESS in 1h 19m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e88308149ea47dbad9b7c208473dd63 : SUCCESS in 46m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a54a27ed08b4b6cbf4a08224a40a3d0 : SUCCESS in 1h 11m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/76a56f1af86f4964952fdfdee2825567 : SUCCESS in 1h 38m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac98f06546dc4d91adc73c7164205653 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"844e09bac94cdb979da259d3b8a2378d58714af1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-15 16:21:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_01_2025_16_15_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"587b07adcbe4e18da4a60cc4053d6e7b13dc83c2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-15 17:43:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/50/939350/1/check/fujitsu-eternus-dx-fc/9dd6c8b : FAILURE in 29m 37s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/50/939350/1/check/fujitsu-eternus-dx-iscsi/12703cc : FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ab7935ce0ffd4e81428aa6325b99da291ea84435","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-15 19:21:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/939350/1/ : FAILURE in 1h 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2cb95e73b033b22f5a415a133c381cb86d3351a0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-29 23:01:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58d43998b6507a42bbe90435e253294b070ea123","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-01-30 14:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3239434002a032ab227b63f638c87feed2460835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 14:08:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33b7094bf656bca67b4d062214ac179e54283730","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:11:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bd8cae4dff24dc9acf9868ea9d9834d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c94d34a2b94b18ac4c45ec212bd00d : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d5ce857850479baf4d038ffcb76c75 : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc267d3b06542f9b090b6a2ba9a19ce : SUCCESS in 11m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/350b4c6f1697451f9bc6729d9c1d301c : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10cffbc31a14667a675f763f29f2def : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/bfba909458e04ad0a465e1203d1919ea : SUCCESS in 33m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c68d4ef47a94dd0a395a55437c5fc1d : FAILURE in 1h 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9068c4f685b44477bf6e3796ea9904bf : SUCCESS in 1h 01m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffeb06fc761347b9a2f66dd21c979b96 : SUCCESS in 18m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d237e675546d42ababd7282c8465a8f2 : SUCCESS in 1h 11m 17s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04c5ec3da1e24e48aaa61f340a8e9833 : SUCCESS in 1h 46m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18b2ce4285dd4695856eb9eeda985650 : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"261b9656185a4d3e8c56d6bf2fc6961ed3db52a0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-30 17:30:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3503507d0fa456f65631193a2524bf0b0797f9dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 17:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca407f8919864c5c93194e54335c0e09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/418c74f8006040908d9ffa4bcee65c55 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9f5fc9de24a40a18c13c953af506a80 : SUCCESS in 20m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/781c9b600e1847f19d262e233a4108df : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35936fbc0e574217bbb87e5f24cbb4ae : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd6a4037bd4e9f08fa3bc1ab41120c3d260be0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 20:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4a137469c1f431cbdfb7ead62eb3e6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb3f427b2e34af9bf5b60010418d0eb : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eecda8a5f6a6439794c0cda643035e22 : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a062cc66a95c489495ef4f587d6f0c32 : SUCCESS in 12m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f788beb861b4a38b4fc716702445cd3 : SUCCESS in 14m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f8cd630b3e74703aeb972c73139419e : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f6b6f972b5455eb7545821f4e2e39a : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/3e61871b75c148f1a64880429bd2c591 : SUCCESS in 58m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ceb93e8471c0497189204dd3fbb34438 : SUCCESS in 1h 25m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce71d6bec792438ba165cd34f75321ec : SUCCESS in 34m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caf46461345642088bd2fe89fd005243 : SUCCESS in 14m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5084107296d40daab16a34338e8300f : TIMED_OUT in 41m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0ab2e9466dba4192967e57befdf72390 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51c1d3b5b31748dca3589628926f4cf5 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/296f39b5f99b41709b25c2dd832c6aad : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc4200e0e51842958daf0940bf0440f8 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/044e5100769648188750e8925d38178b : SUCCESS in 36m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe2f5a80b8f34594babd927980913977 : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a72a03ee86fb4bbdb99679ec60eec718 : SUCCESS in 1h 27m 58s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fb78f53a3004cda919e67acaf845f43 : SUCCESS in 1h 47m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fb85e31627854af0869ff615efe6c23e : TIMED_OUT in 3h 07m 18s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4d936fe246c479c865005cdeeca97f2 : SUCCESS in 1h 10m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eabd6ee75f3143d4b2a2f0696302cdf2 : SUCCESS in 1h 22m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/176c33cfe022489ba273fd04f49789d5 : SUCCESS in 26m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/481f7090c76142ac906b8093a96bc52e : SUCCESS in 1h 32m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/22d46dab7f234e3a93342d29e2c2ec74 : SUCCESS in 1h 37m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33f0a5bf561e44708f6515ba3f6283e3 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d87ca919dcb50c112597629e13797e61cabe63eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 20:59:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcffc4b81d31d117d12b8167ad700aaa0e527fd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-30 21:04:00.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/30_01_2025_20_59_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"2408c7a1979fc4fec7bbe9024fa713639191eb02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 22:52:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25aa9dfaf77e46e2bd0ba71ce61a1477\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f5d9f63eed4853a1a548549ea35ae4 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2354e1de46a642608fe46a6fe8949967 : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/507fa6b5cd914095bf9e6cc336a2ac69 : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88e11e4146af4112a67565f846b22e5f : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4762a73741fc458c967445726a4222f5 : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/4591dfa77b28477aafea053f51ced986 : FAILURE in 34m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/683a1c2b215f48588c72c06f3d2d38e1 : SUCCESS in 1h 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb05f8159ec74690af96f0d2f9aff017 : SUCCESS in 1h 02m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e9bda44c3454cd0a160c14be6447bc6 : SUCCESS in 18m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e1154ed7e8ef4251b3ebed7fb071e621 : SUCCESS in 1h 15m 31s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c96d515fc594189a2e9222bdaa28c1c : SUCCESS in 1h 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/688679ad006a4464a435def3c1281757 : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"83d5a905c30d77ae13c3267051901436ce8fd081","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-31 00:15:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bff7e589815210e7d27dd59f0b29d2a4262702b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 00:42:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed60e2f9523a46039cb149a9f6933944\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78024772c5c6421d86d82ab8e3ebaa41 : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8e18d96edae4fe299323e596a47655c : SUCCESS in 20m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b75d58dd90534fb58250cda14f0915d5 : SUCCESS in 24m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/69748a5ea7af4124b24df9bc3c495ac8 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae0e02cfa661492ffca2126808077d789005ec40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 02:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2783499d8f5404a83b4a41acf01d9f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5bf4e32e22845a183965f4be9e785f1 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de60912ac13f40f49c439b016e2c4d20 : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69a32d854854429b8e9b48464c412bf8 : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f15624bd5cd44ddd811f8771769a3ead : SUCCESS in 9m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8ec73a0b02248aaa777ca2d4a20891b : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df12e7a146254ad284c69a619dd3781e : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/c70c37da71904635b1f21d5472cf98eb : SUCCESS in 38m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf5f3a5c9fff4206be5678afff6bc75f : SUCCESS in 1h 07m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc3c2817b31841dab775933ef9375ec3 : SUCCESS in 58m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd743545d67349b7a8441d9625753c0a : SUCCESS in 12m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f667af02fca4cf6a05b643e7a5b7a81 : SUCCESS in 18m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3302a73167bb43cdabb94808bd286759 : SUCCESS in 3m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a7a8824e5d843058a5831368c4ec11b : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4335e5f4195043b39283b2509d690506 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d7b040f0ee742cb898c87fda782463d : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4419fdef7d3544d99acb7af54848795c : SUCCESS in 44m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/875e9f39033849d0875f29754c2ebc90 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbbf3654248b46b1a62fa4f42d1f069f : SUCCESS in 1h 54m 35s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/639023d69f5748ddaa5e282b05f78f04 : SUCCESS in 2h 20m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/21a01f77dccf4ae88115efb5e730574b : SUCCESS in 1h 50m 43s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a42fe3a71e147ddbbcfff55202e31cc : SUCCESS in 1h 12m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0e31200b30ba46ff87823302a585ec52 : SUCCESS in 2h 04m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a4765fdc9384034abe7317c18747e01 : SUCCESS in 46m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/676c3e2019244518b984af93e93b2dcf : SUCCESS in 1h 23m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/462f16accaa14a5c99fea4dec2aa29f4 : POST_FAILURE in 2h 10m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5046021b37184b4a96c99d8d8c674035 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2dd2a222dc78844d35d28004b15fe681dc55d741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 02:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34e822a7cfd1873fd9653ed483ef2f2bb5850d23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-31 02:48:16.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2025_02_42_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a1ee539a70bb75e07497d3ccf873f165c3df9d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 05:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5485daa40c5421aacdf63b6efab2ddf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d627855ec343ae853f1cb389ed9c02 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81b2d67b3c77458894ef07679137e6ae : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa5eeb30ac3e4595b007fe1a7099bcb2 : SUCCESS in 12m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d0f11993db645029ae4c86fd26546f1 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dde85f2252349239c776f013d1d4080 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/019c6e9371ac4c8da40b9e1444ff521f : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05590616427647beb1120be89e5a0abd : SUCCESS in 1h 16m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/091994aa0ae9434bbe2aac7c20063acd : SUCCESS in 58m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27cfc1907eec4611ac3e6677b177ac6e : SUCCESS in 15m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d71d69613eff4d30a7e0993eb8a3ae3e : SUCCESS in 1h 03m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7433a4bb6ee245fa999acf40c90f5c1c : SUCCESS in 2h 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/489ac6af8aeb472db631d54bd6b0eea0 : SUCCESS in 44m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4a5c65e5dea1758e8204e198ff877f8e2dabb7af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 05:04:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aead13c714b611f50e61959fe8215d32be8308d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 05:05:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8583a402efce47dbbfd48f4c65efe1cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f227146eb3444cdb7534a5a85ae3221 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/755b398054a04ab2ae27f485ba47aebd : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0480073b9ac64f6897fd6bcea72a8d1cc6d474bd","revisions":{"0480073b9ac64f6897fd6bcea72a8d1cc6d474bd":{"kind":"REWORK","_number":1,"created":"2025-01-15 14:14:58.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/50/939350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/939350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/939350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/939350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/939350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/939350/1"}}},"commit":{"parents":[{"commit":"a770f720860d1603e4501b02960e3d7f13c72f95","subject":"Fix \"signature_verified\" metadata propagation to images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a770f720860d1603e4501b02960e3d7f13c72f95"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-09-15 00:00:31.000000000","tz":0},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2025-01-15 14:14:47.000000000","tz":60},"subject":"NFS update volume attachment format during volume snapshot","message":"NFS update volume attachment format during volume snapshot\n\nDuring a NFS volume snapshot of an attached volume, a QCOW2 snapshot is\ncreated and is made the active volume for the instance. The associated\nvolume attachment is however not updated, resulting in an unbootable\ninstance after a stop/start.\n\nThis adds an update to the volume format and attachment connection_info\nformat during the snapshot.\n\nCloses-Bug: #1989514\n\nChange-Id: I3c638dab2f0b5a2128eaeeea533450d7cdfa1577\n(cherry picked from commit 56abc9d5b9a07ca31791e385d93ca5b56b6bd74e)\n(cherry picked from commit 9d1a7de850ad06e9f8e242ecfeb070da22c688c4)\n(cherry picked from commit a54442aef0860876db85b29f8df1b2c1245aecd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0480073b9ac64f6897fd6bcea72a8d1cc6d474bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0480073b9ac64f6897fd6bcea72a8d1cc6d474bd"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
