)]}'
{"id":"openstack%2Fcinder~939351","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I3c638dab2f0b5a2128eaeeea533450d7cdfa1577","project":"openstack/cinder","branch":"stable/2024.1","topic":"bug/1989514","attention_set":{},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-01-29 16:01:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c638dab2f0b5a2128eaeeea533450d7cdfa1577","subject":"NFS update volume attachment format during volume snapshot","status":"MERGED","created":"2025-01-15 14:15:16.000000000","updated":"2025-01-29 16:03:26.000000000","submitted":"2025-01-29 16:01:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939351-bug/1989514","meta_rev_id":"79a06c03588286a8a8d3501ffdb2b87ed502b95d","_number":939351,"virtual_id_number":939351,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-29 16:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-01-28 22:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-01-29 14:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-01-29 14:03:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 14:15:16.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2025-01-15 14:31:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-15 14:50:46.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-01-15 14:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 15:10:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-01-15 15:55:28.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-01-15 15:59:58.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 16:01:10.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 16:09:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-15 16:53:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 18:13:59.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-23 23:06:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-01-28 22:50:38.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-01-29 14:03:47.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"f427c149d42526d69b8fde5e01f35dcffcaddb16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 14:15:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c968b0c7fae4c4c2258eb425094d6de93e6900","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-15 14:31:05.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a57514d558fd4070a700698af65f9243\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/893755be0c6b43aea93c9bb2cbfded4d : FAILURE in 6m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2170f282b8b498492a78dc23bf9f7a6 : FAILURE in 6m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d443bc664cf749769db16629a1141f3d : FAILURE in 7m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3162f4e6f1d34e58a9dbee90939730bf : FAILURE in 6m 56s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:s... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"b5b2ec1c5a6057032367ee9c10cb20d11edc847a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 14:50:46.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1220/","accounts_in_message":[],"_revision_number":1},{"id":"bcf650729c60fbe7926631d1ea5b0a567e94d9e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 14:52:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7eb0aa0a1404ecfa115d2b460531c5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/824efa9790834b7b90396c66d41ccc6b : FAILURE in 24m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3182d752259f48eb92b776439716b30a : SUCCESS in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6116e70fbaa02b2bba8e47c8e8aff569481a58a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-01-15 15:03:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939351/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939351/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"2668a5694d60ad8c2986f06031c71a7f4c906831","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 15:10:09.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/243/.","accounts_in_message":[],"_revision_number":1},{"id":"ce587a42904a5882b5615d590593662ff5067f3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-15 15:27:22.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/231/.","accounts_in_message":[],"_revision_number":1},{"id":"c1fc5e73f368a98961d12581c5bb739111381f23","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-01-15 15:55:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939351 : FAILURE in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7a93cd1a6577e23524f883480bb7e5aeb44d386e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-15 15:59:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/51/939351/1/ : FAILURE in 1h 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3e50dcfc526a0f05c5d89623b6bc9cc917a4007a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-15 16:01:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939351 : FAILURE in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"42d0fa59a3318d6f4641849b07ca4ea2962954da","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-15 16:09:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939351 : FAILURE in 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ba7ea3e755f2e2c67c59f988e80b8d202585b24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 16:53:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/757abf1f54d54f41987cee13dee35c4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48891054013a4145a35dc536555f38b8 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20748a5cfab740e984532388e1fbad54 : SUCCESS in 12m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2be395ccf34649b5818a0d404ff04851 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3c6bef32914fd78e6d1baebdab403c : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/abafed5fedff4114a82477b8932ed093 : SUCCESS in 56m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/77ba75977a094a94a84405e960cf058c : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ded3d5f14367480e84aa5949fb92fe2c : FAILURE in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbbd85c3a9714930b39c069fd776b49a : SUCCESS in 1h 02m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbc2c189f2504fa08086993cac40fc30 : SUCCESS in 16m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22c60d2ce9254a77b29fc8b1177fe6be : SUCCESS in 17m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6624025f61d744fcb8b20012173e6190 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/560b9067275440feb7c4b95ed3945b44 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb8ae3b0bdd649a7bc8652b504053edc : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/751a3cdf92fa426ca26a3f35b2ae4f91 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1e6e963a874d4778944b0694939403cb : SUCCESS in 43m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/365285caf2ab41efba8ab425105333e9 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5d84445ba7148d28ab8f8bda08d0923 : SUCCESS in 1h 33m 13s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/536f16b3cfe74c868fdd3ac98fc56baf : SUCCESS in 2h 24m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/117d90dad38c4f5493954d4f2f55bdfb : FAILURE in 1h 49m 39s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/51769ad6ab2b469eb78056c04bf7d1d1 : FAILURE in 59m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b588cc4277f04bedb129aace776bc4d3 : SUCCESS in 1h 30m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f3542a303eb948cdbf6d2bfc88abfa3b : SUCCESS in 34m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c117f035a914476dac80634f9c9e771e : SUCCESS in 1h 26m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a20c46c0a31648c3aeaf5ac294d16088 : POST_FAILURE in 2h 21m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e27aca012b7421c80c5edc1b21d5a1a : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":1},{"id":"06be3ed42a8ea5aac869da374f7d7b57c8ff814b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2025-01-15 18:13:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/51/939351/1/check/fujitsu-eternus-dx-fc/feee608 : FAILURE in 30m 05s\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/51/939351/1/check/fujitsu-eternus-dx-iscsi/d88781a : FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fa4fee8eeb25c133e202dc68d83cb0f37bcbe06c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-23 23:06:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ed9ce69f9ed58431da7af587369cb71d9e876f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 00:52:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45ad3cac8a14ae6b4fcb828630ff36b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/531945fe133a4af88840794dd67af091 : FAILURE in 27m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7c0955047af41f981b786a020d79a94 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccdaaa54eff49665bbdc0c6f6cc6d5078616c897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 00:55:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aa01da449b64504bbead7f361414b9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2e88050dae4cbcbc2db4d45671ca68 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcb58d43b92642dabe66f090d9812913 : SUCCESS in 8m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b244201b86c493ba02f71243510cc5c : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd441c866fe41bdbb13128abf5f763b : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/0a80c77dbf44467cbdbb15b20e52b7c5 : SUCCESS in 1h 04m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ef13dbd71aec4b049fb67a6e41c29ef3 : SUCCESS in 32m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0c2b20e23034e4c816e3137cc57bf5e : SUCCESS in 1h 37m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d99f86547e54560be85949ed95246de : SUCCESS in 36m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b93211512853471ea31bed6ff012fafd : SUCCESS in 18m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f72214a38564ef18eb9dc8f629b711e : SUCCESS in 11m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cea3f6820e74734b9b61cfdc36a7bcb : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/727f6b5b72d04264adecbd4a192cbe1f : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e641d9903e84ffd892c5773bc400f64 : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8e31d86dae3433d9010b799f18f3f92 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/403f289c5b6842d59e43ac6bd847e9d0 : SUCCESS in 32m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d72f594fa7da45ddb1e70f6fc7559c7d : SUCCESS in 4m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ebf6ed250e8a4b67b209a2790e125031 : SUCCESS in 1h 42m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0d64492b38a475e93faa156172febb0 : FAILURE in 1h 36m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/87aa8b65fee440ec83796037628e8c28 : SUCCESS in 1h 47m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93f42ffacf3945dd8d15d6a1aa295262 : SUCCESS in 1h 05m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df2e08644a4349c793633df826d1008a : SUCCESS in 1h 32m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ccc9cefa5b614c4488566c4cb6b0718f : SUCCESS in 34m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d85b1780e374647a12381e770e43e19 : SUCCESS in 1h 29m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/af6083ceeefe4fb5af425f5274d34d3f : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee54f31158104c5689e15f1095a2e38d : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8f4a4df6fdfba17a2bd2581b6af9ffe87904a030","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-24 10:13:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d5a9a66b9cdaa13814e63d8be158d49b950417b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 10:49:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc4615c12b04d7da15780599bf32035\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4cc369be63649e199da2eae68ff50fb : FAILURE in 26m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1687870493a4c7aa53478f13cf7e61c : SUCCESS in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53f3c55c26a4824b06c7cf6725ec01c9e352098c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:56:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe35e84c053446bc93ef19d656095974\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826fa2c42ffc4250809021d26abab277 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4d9f19c954e43deb8b25e73c354b864 : SUCCESS in 18m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c505a56f2e5a4c319684dca39080840f : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c74abc101624a9e94634137e2916603 : SUCCESS in 26m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/0e23fbe5203a45e9a37c8e1536cfc90e : SUCCESS in 1h 11m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cce80825c3fb43feb5260c9ad54c04b2 : SUCCESS in 55m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df975ac1e3484fcfa6150d038f9bcda1 : SUCCESS in 2h 03m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/369d08f524c3498a84de375636669ac1 : SUCCESS in 1h 04m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa7d90b751d84302955632f05a82d6a8 : SUCCESS in 16m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf72f9535b7a4887a6e62638051f46a0 : SUCCESS in 24m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e2cb8e5a8664af38a58c3112494d5a1 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b4c341736024dd9902eb406dbd38670 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/366011690f7a4bd3b0a555bf3b01e827 : SUCCESS in 13m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30a481ca35fc4ab7aaccdbf06585be92 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e99833aee088466cab9d8a0d3e3329c5 : SUCCESS in 54m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f92f27f949dc4f32b0fe9d8e7624d836 : SUCCESS in 13m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/80df6b5a222d4b429aebf5d14f3aa678 : SUCCESS in 1h 46m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1b917962b904509b7ac2238e5496daa : FAILURE in 2h 33m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2ee26b868e3f45228d29a174fd1a57b8 : SUCCESS in 1h 53m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9167c9ea526d41e2b472a058a6d6c3cb : SUCCESS in 1h 06m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8bedb70658bc43269251e05c0aa75ec3 : SUCCESS in 1h 44m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b4a81fcd03ff490cbdd180da2d98fc6e : SUCCESS in 35m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df4a9cc8de834b4a817611860dc88f25 : SUCCESS in 1h 22m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/35c03cd5d1aa412b8aba8cebafa0da23 : SUCCESS in 1h 46m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e76adcd6cfbe4e68b429bbc82140f8c7 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3b6ec792a16a0d9450e8de333b8b8585e24858ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-27 15:36:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c14f78824c8a99d866df13370c971ccaa3aa001","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 16:27:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24bc1f99d928477899c6ddf1aab0fad8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5b77ab3e8384235a19ba607e81986db : FAILURE in 24m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/970a777716fd498fab5cab8a0a9fc58f : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6481c48c4ad8a39c91ba94d7fa62e39959a36b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 18:07:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f045e1142440cf90789e63c7ce61c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105c007a865a4562bf2be8def2b7310d : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc87687e282540d7952911b52285baca : SUCCESS in 11m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/988045a12222484bb691dec5acf66b20 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef74ddb855a141bfb34067157d9d06d4 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/ff13523727744f13804fd7548b9d2285 : SUCCESS in 1h 10m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f788bc5fb904d8aa693e31ab9c6dc80 : SUCCESS in 1h 07m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/104353f1c42948cb9107ea05137af4ec : SUCCESS in 1h 31m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6f08435ac7e43d0bf922d4ea4ecda3a : SUCCESS in 1h 50m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/328e847b3cc64a6589b7002179d16a1b : SUCCESS in 23m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f83d8b94179f41dca9eca0e0a16f4296 : SUCCESS in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6fbeb562fee5488ba4c8204b8a8133b4 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bbb13cc6667c4379b58bf918fa2ea400 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5967bea8d61940a0b447bee192ada772 : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6dca64f6b60941c2b5d2c8e9f606db81 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b88144fe356e43b3947ea66435e310df : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22e391ff6e6d498698d0ee2746509707 : SUCCESS in 5m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70a2d571428648be935074a53242521e : SUCCESS in 2h 16m 41s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3d4eea1833f40c19d443871c19d82a4 : SUCCESS in 2h 08m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8cdf6e45dc6747258bdddda39f385596 : SUCCESS in 1h 57m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/552224fac2954447ac98de77ef2f029c : SUCCESS in 1h 17m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82cd2ca8482f4c75ab3544178e7d3d30 : SUCCESS in 1h 27m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ff4e69210ed14eaea880b6c533af5ad5 : SUCCESS in 47m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93721c84809548738c7cc4f451d81da4 : SUCCESS in 1h 23m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1bfe14ce5a194bc4839c968f4262427e : SUCCESS in 1h 35m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34ed1394871a4012874c974a6414003f : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e60860c859aea3f5ae554b2744ce0cd41f3a3cac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-28 22:50:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6ed1bc5a12a58f551d49fc0a094a083fca8315d","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-01-29 14:03:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01f7ee405beec282a41a04da30f097f82134f96a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 14:04:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"770d31cedc2250b6976c221ff283518f91fdc77c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 16:01:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a82618a8aaf4c489ed9fe59f215dfe9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbdf77ce6a02482f93022f8c2a91e278 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/801fb042e60c4cc5b4be0894c7623df6 : SUCCESS in 12m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a668232a4b0e4640832780ba6d86d7ba : SUCCESS in 19m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217cda4418244592a9af9483739e4cd3 : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/598c707152b34b6cbc115b31cb230d46 : SUCCESS in 47m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3eb3fce719944b5cb8cbbdc300719586 : SUCCESS in 49m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1199e23e0f3040c195628997a13769e3 : SUCCESS in 1h 23m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84873c297c4a43edb16673392761ed25 : SUCCESS in 1h 30m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c393e9eb748425887ddaad51a84386d : SUCCESS in 14m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64306707998f4dfd90f4013f24646771 : SUCCESS in 1h 08m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51199a74ea7243cba8fb02796d2e1a9f : SUCCESS in 1h 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe352e9338bf44d1a4adf6af6477c02b : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6f606ee3347ff7a1ba23d20c25979c3b0af13844","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 16:01:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79a06c03588286a8a8d3501ffdb2b87ed502b95d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 16:03:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c115d5a7895c47619db0e2a3f0179111\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66a0f1cca02475da239f09a983e8221 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4a199acaf7f41478ea68d390eae5d79 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a54442aef0860876db85b29f8df1b2c1245aecd9","revisions":{"a54442aef0860876db85b29f8df1b2c1245aecd9":{"kind":"REWORK","_number":1,"created":"2025-01-15 14:15:16.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/51/939351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/939351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/939351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/939351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/939351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/939351/1"}}},"commit":{"parents":[{"commit":"eeb592af5e75855fa85fa9c6dec3abea8fe86740","subject":"Merge \"Fix \"cinder-manage quota check\"\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeb592af5e75855fa85fa9c6dec3abea8fe86740"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-09-15 00:00:31.000000000","tz":0},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2025-01-15 14:14:33.000000000","tz":60},"subject":"NFS update volume attachment format during volume snapshot","message":"NFS update volume attachment format during volume snapshot\n\nDuring a NFS volume snapshot of an attached volume, a QCOW2 snapshot is\ncreated and is made the active volume for the instance. The associated\nvolume attachment is however not updated, resulting in an unbootable\ninstance after a stop/start.\n\nThis adds an update to the volume format and attachment connection_info\nformat during the snapshot.\n\nCloses-Bug: #1989514\n\nChange-Id: I3c638dab2f0b5a2128eaeeea533450d7cdfa1577\n(cherry picked from commit 56abc9d5b9a07ca31791e385d93ca5b56b6bd74e)\n(cherry picked from commit 9d1a7de850ad06e9f8e242ecfeb070da22c688c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a54442aef0860876db85b29f8df1b2c1245aecd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a54442aef0860876db85b29f8df1b2c1245aecd9"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
