)]}'
{"id":"openstack%2Fcinder~939570","triplet_id":"openstack%2Fcinder~master~Ib1045dddf3f82cb983111882582efbba56d59c97","project":"openstack/cinder","branch":"master","topic":"doc/multi-tenancy","attention_set":{},"removed_from_attention_set":{"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-02-03 06:50:30.000000000","reason":"removed on reply"},"36569":{"account":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"last_update":"2025-02-24 21:40:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib1045dddf3f82cb983111882582efbba56d59c97","subject":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality","status":"ABANDONED","created":"2025-01-18 05:30:13.000000000","updated":"2025-02-24 21:40:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dfb2d6d8a82f2dbc315d5ad93c9946e5e7012f0","_number":939570,"virtual_id_number":939570,"owner":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-18 22:59:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-18 05:41:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-18 06:20:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 06:40:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-18 08:49:04.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 09:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2025-01-18 09:16:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-01-18 15:10:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-01-18 23:01:24.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-01-18 23:01:30.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-01-18 23:02:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-01-18 23:02:20.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-01-18 23:05:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-01-20 23:10:27.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-03 06:50:06.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-02-03 10:57:42.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"}],"messages":[{"id":"559633355bc4f63a38fc3b24464ca25b71ed511a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 05:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3a77ac9560661ee4aeae3f24d0f3e1327f1a574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 05:33:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d1f4623084854ea96b77ba79b90c9d2eb095c86e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 05:34:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0a7b71dc6e26f7e66056b460e23a968678753b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 05:41:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/898ed4787fcc4208b4639aae8a0c2246\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3dd31fb8ed4ed3b3c910cbbf16ddf2 : FAILURE in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92cd9204d16440d2b9b7df9b9d2bea22 : FAILURE in 5m 13s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/002406286ebe46e194e74f5c9af93f61 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"0e0d8d020c83da532ecca804125bd415d7087b20","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 06:16:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0d567b4a1864a55cfe5a91afb7a129cbf51f725","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-18 06:20:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939570 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":3},{"id":"294c38892cb957311d00c1b10656d9d8e0e91aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 06:22:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48181656dea144e2a9675baf0d95fde2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8dccc1ff4543b482fffef6ee364f4c : FAILURE in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08b1b7424e441a9bbb4dfff0fd93e4d : FAILURE in 5m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51a782d7674242689c148ce316c3b54d : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8128d8f5277b632130180166944d181d4e1197f2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-18 06:40:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939570 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a20e65da6da3c28ac5a5afe07f1e410e0a67bc27","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-18 08:49:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/939570/3/ : FAILURE in 1h 36m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1a75674949a50ba086d80d78e04217f290209f88","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-18 09:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/1/fc : SUCCESS in 1h 06m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/1/iscsi : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"20577b485f9bc7642a695c6603c658af564e76c5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-18 09:16:44.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-1/FC/ : UNSTABLE in 1m 27s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-1/iSCSI/ : FAILURE in 3h 46m 9s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"90d79b97cbce5ffa70c2d77a17c1707db4e2f5f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-18 14:36:19.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-2/FC/ : FAILURE in 9h 2m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-2/iSCSI/ : FAILURE in 3h 18m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ba3265ba6adb6a1914a2e1d5198c8656b7afb1ce","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-18 15:10:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/3/20250118195133/NEC-V-ISCSI : SUCCESS in 4h 05m 51s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/3/20250118195133/NEC-V-FC : SUCCESS in 4h 19m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d6fe1f6e507a16180004e19ef7fd727378c956ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 22:40:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"49748de7f236148a8627f77c10b9b5b4b0f540d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-01-18 22:42:18.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"591913f106047767cffd9399bf3899f516790caa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 22:59:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b6a122f540a4724aab5c54139c59b15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf00362ff2c4305931bff9a16a0a0ca : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71df3a01fcb46f582717ddb4b83f10c : SUCCESS in 11m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/29f4fa45fe6644c28e37545ce4682602 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"222252810d239797ae4c6c067dad242b98cf7fde","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-18 23:01:24.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"22da5c02dea9fe23a47331e01cd56592575fba74","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-18 23:01:30.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"b3ff8d0fc19cc26d6c27b5c5dca5d91ce338c6e8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-18 23:02:11.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"5777d086d447ee0db17272600dbe3aa56c48c2fd","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-18 23:02:20.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"a2d8bd235daeaff1b970b23f132aee34be8dc663","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-18 23:05:41.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_01_2025_22_59_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"0d12f73e8b3a59f3b035815a203d26628a4996a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-18 23:28:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939570 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":5},{"id":"acaa1f86d74bf08b2cbd0d8095fd1f17bc3fd4a7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-18 23:46:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939570 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":5},{"id":"5b83b5840db9a253deaaa0d107782e26fd43dc82","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-19 00:32:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/939570/5/ : FAILURE in 1h 42m 21s","accounts_in_message":[],"_revision_number":5},{"id":"aa89a2c0f564bfc69adb77fa5e6726ff839169a5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-19 02:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/4/fc : SUCCESS in 1h 07m 28s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/4/iscsi : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":4},{"id":"082579084da5a7d897f591f4460fdd236be0c6f9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-01-19 02:08:43.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/5/fc : SUCCESS in 1h 03m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/939570/5/iscsi : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":5},{"id":"0158519b7aef5ad60756187cb9667082aacc1f06","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-19 02:53:55.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-5/FC/ : UNSTABLE in 6m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-5/iSCSI/ : FAILURE in 4h 11m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"c6ae5675e43e236fae31111385b6f2327f3cebc1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-19 07:09:53.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-4/FC/ : FAILURE in 8h 28m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-939570-4/iSCSI/ : FAILURE in 3h 44m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"95482e4a2e323f33609514680d645c9a38e23d82","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-19 15:34:52.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/5/20250119155301/NEC-V-ISCSI : SUCCESS in 4h 03m 51s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/5/20250119155301/NEC-V-FC : FAILURE in 8h 41m 48s","accounts_in_message":[],"_revision_number":5},{"id":"7ca8c3f5e5f7944b17e9b1658a6bc0cb16f15ba2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-20 23:10:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/5/20250120235931/NEC-FC : SUCCESS in 5h 49m 50s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/70/939570/5/20250120235931/NEC-ISCSI : SUCCESS in 8h 10m 54s","accounts_in_message":[],"_revision_number":5},{"id":"71447c7eb6dda23696d61ed93897d261513153c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:06.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1233/","accounts_in_message":[],"_revision_number":1},{"id":"95919d756c5b293d67de9918996d1731bdee7f8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"41829cae3250f44bf1677b31854f08c70c61591d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:08.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1234/","accounts_in_message":[],"_revision_number":2},{"id":"708003add930d6ce326f161cff5808834e9b8409","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:10.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"a40bb966c60feb26061daa405ecf0d4f3affce3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:11.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1235/","accounts_in_message":[],"_revision_number":3},{"id":"9e1eed0e0f335c7928d4485f8d12f1c5d3708e94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:13.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"8e5aa26dc5e134147cacd8c876b7a640fcc70dc1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:29.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1240/","accounts_in_message":[],"_revision_number":4},{"id":"407c18673e60e61b02ea1c61d0f49481927b08a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:30.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"4bc0bed76b9c7c4151ea283f8db238f44cb47153","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:32.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1241/","accounts_in_message":[],"_revision_number":5},{"id":"d19766a7d056b437fc50bec9b31f0495fce8cd10","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:33.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939570/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"87897aa4d37e274315ed20e74dbb2c925174dfc7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 10:57:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939570 : SUCCESS in 39m 51s","accounts_in_message":[],"_revision_number":5},{"id":"3dfb2d6d8a82f2dbc315d5ad93c9946e5e7012f0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2025-02-24 21:40:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4c37cfe0d6103a94aa0641cb496705e9014b3b80","revisions":{"25be4138c56ceaac113c90dc9c29054573c5f6fc":{"kind":"REWORK","_number":1,"created":"2025-01-18 05:30:13.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/70/939570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/939570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/939570/1"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"subject":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality","message":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality\n\nChange-Id: Ib1045dddf3f82cb983111882582efbba56d59c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25be4138c56ceaac113c90dc9c29054573c5f6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25be4138c56ceaac113c90dc9c29054573c5f6fc"}]},"branch":"refs/heads/master"},"e0a049f57c1a4ba401678fde7ed8b1531e0ee8b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-18 05:33:56.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/70/939570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/939570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/939570/2"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:33:56.000000000","tz":0},"subject":"Update Lightbits\u0027 Cinder driver documentations","message":"Update Lightbits\u0027 Cinder driver documentations\n\nUpdate the documentation regarding Lightbits\u0027 multi-tenancy functionality.\n\nChange-Id: Ib1045dddf3f82cb983111882582efbba56d59c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0a049f57c1a4ba401678fde7ed8b1531e0ee8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0a049f57c1a4ba401678fde7ed8b1531e0ee8b2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1313722bd6082831d51b33fa475019a906c4e60d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-18 05:34:47.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/70/939570/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/939570/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/939570/3"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:34:47.000000000","tz":0},"subject":"Update Lightbits\u0027 Cinder driver documentations regarding multi-tenancy","message":"Update Lightbits\u0027 Cinder driver documentations regarding multi-tenancy\n\nUpdate the documentation regarding Lightbits\u0027 multi-tenancy functionality.\n\nChange-Id: Ib1045dddf3f82cb983111882582efbba56d59c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1313722bd6082831d51b33fa475019a906c4e60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1313722bd6082831d51b33fa475019a906c4e60d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"888b8ee5a2f8302bd7cd91f73e4b18a37377b7c8":{"kind":"REWORK","_number":4,"created":"2025-01-18 22:40:51.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/70/939570/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/939570/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/939570/4"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 22:39:10.000000000","tz":-480},"subject":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality   - Removed exstra space","message":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality\n  - Removed exstra space\n\nChange-Id: Ib1045dddf3f82cb983111882582efbba56d59c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/888b8ee5a2f8302bd7cd91f73e4b18a37377b7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/888b8ee5a2f8302bd7cd91f73e4b18a37377b7c8"}]},"branch":"refs/heads/master"},"4c37cfe0d6103a94aa0641cb496705e9014b3b80":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-18 22:42:18.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/70/939570/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/939570/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/939570/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/939570/5"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 05:29:25.000000000","tz":-480},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2025-01-18 22:42:18.000000000","tz":0},"subject":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality","message":"Update the documentation regarding Lightbits\u0027 multi-tenancy functionality\n\n  - Also removed exstra space\n\nChange-Id: Ib1045dddf3f82cb983111882582efbba56d59c97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c37cfe0d6103a94aa0641cb496705e9014b3b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c37cfe0d6103a94aa0641cb496705e9014b3b80"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
