)]}'
{"id":"openstack%2Fcinder~939572","triplet_id":"openstack%2Fcinder~master~I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885","project":"openstack/cinder","branch":"master","topic":"remove-windows","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-02-16 21:31:35.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-16 14:31:18.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-02-16 21:31:35.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-16 21:31:35.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-16 19:51:33.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-16 14:32:49.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885","subject":"Remove Windows drivers","status":"MERGED","created":"2025-01-18 05:51:08.000000000","updated":"2026-02-16 23:12:45.000000000","submitted":"2026-02-16 21:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939572-remove-windows","meta_rev_id":"85b9ad6fc3d778f79dc5011f77786cf79d40dc65","_number":939572,"virtual_id_number":939572,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 21:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-02-16 14:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2026-02-16 19:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-16 14:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2026-02-16 19:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-02-16 14:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-18 06:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 07:39:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-18 07:49:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 08:30:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-18 08:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-18 08:49:45.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-18 10:48:24.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-20 03:45:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-02-03 06:50:17.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-02-03 09:33:57.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-02-17 10:54:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-21 14:51:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-02-21 18:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2025-11-20 16:32:32.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-11-20 16:33:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-11-20 16:33:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-11-20 16:33:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-20 17:30:36.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-10 13:58:24.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-11 03:50:04.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2026-02-13 20:43:04.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-15 03:45:33.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-15 05:22:11.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-15 07:30:26.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-16 23:12:45.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"2f3e11d7f399cff1c15da3df48be5458bb7fb6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-18 05:51:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8acb0594ef2e17a0bf60938c0288305133f5705","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 06:14:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5aba4ca110a4bc5b7c1795e840b31e9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0222d7f53208416e9f3480d138240ffe : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0460c9e3e56477f9b1bbd0c5bafbef5 : SUCCESS in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"825cacfca6e52b737b95f58ac8bbc32a43f83da7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-18 07:39:21.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4525fdbbf86c4607afacb64c5984c494\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5edf02d3d014b1ba1a17bf56d30c461 : FAILURE in 42m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc458cab9eb64cb4b29e560bd6c4638a : FAILURE in 1h 33m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4a1e6127319a4c5cb7150397deb38836 : FAILURE in 1h 44m 38s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/609417a12a8b494785f13aeda9c90beb : SUCCESS in 42m 21s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7ca8b59bd91841299f3638b73f6e715b : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/12ffd9d7e1bd4722a8bd1493430cf8ad : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c5eaa2b4c2bd41e7bf4ea1fe22827055 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c855d5508964357b9b7eae61f66af2e : RETRY_LIMIT in 30s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"84b0a7e325965785b4eb56b1a5e168fe9898be55","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-18 07:49:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939572 : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"eae7aef46263b51168eeb645162d8831aefdd38e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-18 08:30:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/acca09e6e4e948809f9e1af5785896da : FAILURE in 1h 13m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a096331177d5486b8e6756c492ac47ce : FAILURE in 1h 23m 17s","accounts_in_message":[],"_revision_number":1},{"id":"666e765f36a63e3c0cceb4b4eb5ffd821c90ec78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 08:30:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06d0eb4739834d5cab5dfac7dcea2c02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81f1c12da2048c299f9acf4c956d158 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95acc734156c485a9adbce1221dc7ab8 : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06b805fdf8d4402c804ce2498f1dbd58 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8626100457bf427d93619ee097f6a651 : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5560f081deeb4cef871f0fa9a4c517fa : SUCCESS in 4m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/020b4f2585924eeabb4741890a2430c8 : SUCCESS in 58m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/884901024e094b9b81b31a0488a0bd87 : SUCCESS in 33m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58d96b48c56a4f1881e8267ae26c83e6 : SUCCESS in 1h 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b41f35cac93d4131a61d60426527c6d9 : FAILURE in 49m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/068e1d0c9e0b418d9383d27a75b7ebe9 : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e774ef7ed02424e813d4dbcb0aeb02a : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/758f44f5c90a43b997bed95cb44052d3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/feb1aa24d57a49f0b91119f936109d01 : SUCCESS in 9m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bcf54d177bcd405996bf6fe236e18986 : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9770065499b34d9f83c4863141f7def1 : FAILURE in 19m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c2900cdafd34e0aa5e1d8bbeac69631 : SUCCESS in 8m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e99596ea3224c65868c880e0dc3bbd7 : SUCCESS in 1h 04m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6729876aaee9492c8baae5a75f6eb34a : FAILURE in 2h 33m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e8503fac4f0423d9ba9e937b6422b17 : SUCCESS in 1h 49m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a9c7fdd1514d45ecbb83c3d888fa2250 : SUCCESS in 2h 04m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1d87372b3344b10a18f382b166505d5 : SUCCESS in 24m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7299f55008914f4e89a11d471a43ba06 : SUCCESS in 1h 09m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a77baeb3e7bb4ae28db52167fa679e0c : FAILURE in 30m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/00290c87c35944268fcac7f5187a0fe3 : SUCCESS in 50m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28d3d54e91a04a088472326bbd55944c : POST_FAILURE in 2h 15m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/05f7671a1b29468b88031d4d919d6da3 : SUCCESS in 1h 55m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5a37c92da76423dbe5d394ae263a23b : FAILURE in 18m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c47061962114dac98c5c3cf5d722a5b : FAILURE in 1h 31m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1760ba3064b404a8dce569df891f6a0 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8ae4b8f08e52b5c9a606c83009795721bde543c3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-18 08:49:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939572 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"585b4a60a30f5b4d5a062e326c678221a38222b4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-18 10:48:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/939572/1/ : FAILURE in 1h 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"07a3fc71cd01b49a2a2662a718781a4b8144a3d6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-01-20 03:45:40.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/1/20250120112304/NEC-V-ISCSI : FAILURE in 50m 06s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/1/20250120112304/NEC-V-FC : FAILURE in 1h 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"30f512fd350b9b0defe39eebb3b6661c4c93bb09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:17.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1237/","accounts_in_message":[],"_revision_number":1},{"id":"a3ea3cdb9da87f308a2e51a94b59d4893c6fbd9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:50:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939572/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939572/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"5a2d77167bfed410d2c04ede8be8e17671a6e969","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-03 09:33:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939572 : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6cbd98ecfb68dcca734864ccf1f0f26ecdcc2e46","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-17 10:54:34.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/1/20250217182528/NEC-ISCSI : FAILURE in 53m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/1/20250217182528/NEC-FC : FAILURE in 1h 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7b37eabe8399ee7fe425c51331ff978db87d875b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-21 14:26:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"efcd31ec2d641ed226974514de4762d80eac4237","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-21 14:26:53.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1348/","accounts_in_message":[],"_revision_number":2},{"id":"83451aa199745812f1594bed8e30825aab444a11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-21 14:26:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939572/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/939572/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"01e8afd5aabadc4a163a291ddda018cc14918ce0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-21 14:51:09.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/375/.","accounts_in_message":[],"_revision_number":2},{"id":"86fdbeee7eb4e75740acc75dffb4468518d5b209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 14:59:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623389812aff477988fe7aa3494c1dc4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5038d9f095324142be5d9d4ce1a52ab1 : SUCCESS in 28m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4372caa58ce40f4ade6190ba60da6cf : FAILURE in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71f1e2d9dcce2a9533f6f72dcdc8ada0ba84433b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-21 15:08:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/939572 : FAILURE in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1a01911527fcac6ddf6a6cc7b78b2b09205d6ec0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-21 15:34:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939572 : FAILURE in 1h 05m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a1bc5bd282a51a8bccba8ae1bf478198a9a1217e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-21 15:48:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/939572 : FAILURE in 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c936dacceb691c0d3b77acf99227fb942b78ec4f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-21 16:00:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d1a80b7bebdf4c11a72c4acf634b3fae\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd0ac4724f884f8c8055d40878364009 : SUCCESS in 55m 35s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f539b86942614a64b3fd7d7d3ce66797 : SUCCESS in 40m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/337199f250c1497b81e92a9ea923ed6c : FAILURE in 1h 33m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa6bca6a1bb348ada534b35258f0dedb : SUCCESS in 43m 16s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/674c487132174299861bac4615351088 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/db2a768792514a4281567db32e234944 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd17bff1606c43579f673632ac83f5ee : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43e4c8ab64d24960a0c7330713de02c8 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"fc14bf1b55cb52fb652da7ed68953b14b00eead3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-21 16:36:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/589890b1307e4a0da81d09549d6b8d77 : SUCCESS in 1h 08m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b138e114f2494387b5a38cb46a1c94c9 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b269a8194c3cba32d28758728a09e93c665ddd0f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-21 16:41:05.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/2/20250222004140/NEC-V-ISCSI : FAILURE in 25m 22s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/2/20250222004140/NEC-V-FC : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0260aa4870ec552ba3fd60bee83743a4a6b21689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:12:54.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc3d5f721c3e4a5eaea80d7f9c197029\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bc8f4488a6468fbe0ccfd524c89b77 : FAILURE in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8c2c7cd8a24255b243cc6fd9103284 : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b9ff012f05642d18830af0f325fefae : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cfab2f0fdd476fa3364de1072120d0 : SUCCESS in 7m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e9e8fcb28e142afaff495258156b27f : SUCCESS in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4730b6c87493415f9dc9e0138aae2ec3 : SUCCESS in 37m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7259901a33ca40cc92b8912b523659a6 : SUCCESS in 1h 07m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e85ef2864094391ac3ecd9e7073fec4 : SUCCESS in 1h 31m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b40473481da541cf909a269f0a20b0c8 : SUCCESS in 1h 06m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26c20ea74a584367a57a68f4875360de : SUCCESS in 16m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f3a4d14290be43d8bdb791eeab5077ef : SUCCESS in 22m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e09b20f412c4032a2107c2540dacbf6 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88a0ce51db054c278cb77389d708ebbc : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/494570e51d78459380bb937ea1302a2a : SUCCESS in 6m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/582ac8682d38403e8ed972bfda1eaef6 : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acb03d96bb6c46e0aac2c0ef42486b03 : FAILURE in 32m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63aaa2fdd8664f0896b3ec9ca67c5120 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/99b2d39a5471495aa449b40ce1ef05b4 : SUCCESS in 1h 58m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5042f43aac48428cae4b5577a2d7d36c : FAILURE in 2h 31m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbb12dae4c3d497684ba88ead113d667 : SUCCESS in 2h 09m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a70d3193b9004d57a06632d1632674fb : FAILURE in 2h 05m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/737e1ebe57e647e1b4ac180cc353685c : SUCCESS in 31m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/39880a82e29e43059608f63bbdffcb20 : SUCCESS in 1h 09m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3b2970bc21842d1b2670fecf6c08d27 : FAILURE in 32m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0542b4a915b74b83b82a2f0b2b623d8f : SUCCESS in 32m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/02915717e7e44fd0ad2e7a5f21f95a3f : SUCCESS in 1h 48m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9b88ae3f3cec471ebe77f21794125051 : POST_FAILURE in 2h 29m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fdfefc1cc889494db84b92cce4cf8f92 : FAILURE in 22m 48s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab9bd080e9584df79e3ed2682fd93533 : SUCCESS in 1h 42m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f934de121e404815b1bede0d9cfb684a : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4ddd6d3e6af60bb36579d968030bdf2d402407a8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-21 18:00:11.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/72/939572/2/fc : SUCCESS in 0h 50m 46s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/72/939572/2/iscsi : SUCCESS in 0h 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f6488b43c364704fa6253d96d2ded71ed75d8122","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-04 16:18:33.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/2/20250304234836/NEC-ISCSI : FAILURE in 53m 39s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/72/939572/2/20250304234836/NEC-FC : FAILURE in 1h 29m 54s","accounts_in_message":[],"_revision_number":2},{"id":"76efdf8db77840e90d39205879121ca6e9f8a8a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:31:55.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cabde49c8fa5db1f96c7b04e5e90f653f27576c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 16:32:04.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 939572,3","accounts_in_message":[],"_revision_number":3},{"id":"f5851151a450573df037c2c9a1a26a669000e0a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 16:32:04.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 939572,3","accounts_in_message":[],"_revision_number":3},{"id":"d9cc00a23e6c893b7c5f5f476f68a87ccf55bf54","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 16:32:18.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 939572,3","accounts_in_message":[],"_revision_number":3},{"id":"bf84baa9c412bfd23dad261e64aa1ba21fe28c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:32:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"97c94ed1d0bea65d0d2c82c37443c4bce78e23cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 16:32:32.000000000","message":"Patch Set 3:\n\nBuild Aborted \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/45/ : ABORTED (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"712da2ecf1132bd7c31eb458e9f42734d709e6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:32:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5e644acd0ea4f42c3b6eb00f0856b471127d5a41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 16:32:36.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"a2af2c619eea35938a76626318ae933410295bfc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 16:32:39.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"3baa88db89caec267496619ec99cd69dd3994bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:32:43.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"920fc60df09c4955edd92ddb31be32f892978f1e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 16:32:45.000000000","message":"Patch Set 5:\n\nBuild Aborted \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/46/ : ABORTED (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"c2619db1fffd139431fb36eebb34219d09710923","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:33:18.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"afb66d0fe86cc761aa99b7d6fe86e00b6187afc4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 16:49:32.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2550/.","accounts_in_message":[],"_revision_number":3},{"id":"0be923fc57cb737c43dbd4cd8b15e47daa51569a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 16:51:42.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/47/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"64f6933df9a72d24e424c7ad6eeeb5c34cb02139","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 16:58:12.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2551/.","accounts_in_message":[],"_revision_number":4},{"id":"82e1f6ff698dae1fc747d9e76533a6c281c38c51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 17:07:14.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2552/.","accounts_in_message":[],"_revision_number":5},{"id":"4b831a2464e568e513fa080f6ae515389eaf9b95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 17:14:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a013563a795444b8b7fc76679d7ea4f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74de6f71016145cb80a011d3868086a5 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5655e4ec4f145b7ad19f758e09e2e8c : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/367fee936f5240409f9cb75e81318d15 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8bb9af0b5dfe8c2c9521ae8e2e62c83e1691508c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 17:16:06.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2553/.","accounts_in_message":[],"_revision_number":6},{"id":"0bb5c45596dc72cfe7554269b742aeea1b9bbcdd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 17:30:36.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":3},{"id":"70a407449b373aa7196c60e0f0e809a4d2f26c24","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 17:31:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":4},{"id":"a2fcaf2cc50e3fe81fb27fe41826b62a78d67f3b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 17:34:16.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":6},{"id":"9b8777b950eb8bb0425a523af33f4e34bf407187","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 17:36:43.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":5},{"id":"be99a0dd5152adcfc0070fd6e12836c37503825b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 18:31:37.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2693/.","accounts_in_message":[],"_revision_number":3},{"id":"7edb3562d5c78f0451a075bed80f6a27a3f204a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 18:40:27.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2694/.","accounts_in_message":[],"_revision_number":4},{"id":"05bf3f99314f87526825831208af82ec071a1e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:43:18.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f35cc7138f6e4af7abc11b65ce29e66d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a10c8e90ee443408cea868cb38489af : FAILURE in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bd2b6d3abc84546861fa9b44c8f90d8 : SUCCESS in 7m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bde598536ac49eeabd95f7cb0e07907 : SUCCESS in 15m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a686a41f747240f08cf848d2d569aaf0 : SUCCESS in 13m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/267bab896635435c87f880f02817557a : SUCCESS in 2m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/d54811c7a820440e92721756edbefe1c : SUCCESS in 57m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4114555fbdda47c4b16c8ea901bac7aa : SUCCESS in 1h 01m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/662cfb97e9e542078a70ec3ed3c42462 : SUCCESS in 1h 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2627360b08d849129497f0d9764daa57 : SUCCESS in 50m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d50e1680994d02a5021dee0890f4d0 : SUCCESS in 19m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca047db30f5e41bd80d8fafa63c02740 : SUCCESS in 20m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22f116316f45404a9ec3ec9244181e31 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eabaf26cf7864cb78a73f6d1255fa17c : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16bd5ff62797431f912171ebf51c0485 : SUCCESS in 5m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b0df16a993f049ff9ad04dc7c67261e9 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15160d4ab21d4ddda8a42595fa48a705 : FAILURE in 22m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/909ef989681446f983e0848103f802a9 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6797dace08494843afa202634202b708 : SUCCESS in 2h 04m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a940caa8ffcb4738b67c304ca62165d1 : FAILURE in 1h 45m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4ab1f460f5b3498fb1b036a13692af06 : SUCCESS in 1h 34m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7db36766186d41a5b8ba622807a5aa79 : SUCCESS in 1h 52m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/76751891c56e45569ff75107f4175d28 : SUCCESS in 22m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/922539c4ee5542ac94071462437806f1 : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8164d4008fdb4e199f1254b06bb7ea02 : FAILURE in 23m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/17ded6a7be044b839114588b2b87f9d1 : SUCCESS in 38m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4aebfd60100142c5873b88c7d22dcf99 : SUCCESS in 1h 05m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/fd3f82c67ab3492aaab519914e4681bf : SUCCESS in 1h 09m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a95bc9cf7da4d31a324ee685b42eb52 : FAILURE in 21m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3630da9191d3438f98175bfb309e17b4 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":6},{"id":"dd47da6017d3cb49cd3a3af96264de3efacf3d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 18:48:47.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"acd4472fdf59885999a00a7a2b3a5781e6301c0d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 18:48:53.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c24b5daf2318f642590339de108463ddb1a451d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 18:49:14.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2695/.","accounts_in_message":[],"_revision_number":5},{"id":"4733ae65f64175d93222354740c6fce414650fe0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 18:57:44.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2554/.","accounts_in_message":[],"_revision_number":7},{"id":"ab8c206975a12fa6e79f7c1c4e82a9758eafbb69","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 18:58:15.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2696/.","accounts_in_message":[],"_revision_number":6},{"id":"78215af54313fd594ff689f32e05e3ccc4b3c674","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 19:07:28.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2697/.","accounts_in_message":[],"_revision_number":7},{"id":"67140b82b5262eb6f025a3155076d02bdba1679d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 19:08:14.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/48/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"64b69312309f13fe5fc25f2a2c2f0b6fe8bcb67a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 19:32:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33ae86d38191428f86a18664d6b32603\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e73b5a203694828ac16a298e2a784fc : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7217a40c67b4753bc747d963a8bc2ab : NODE_FAILURE Node(set) request cc902b6c8e2f446b83db1c8bcb4ce02f failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b87496628b0d479aba7822da25e9af23 : NODE_FAILURE Node(set) request 5e5efd2df32241da8278abc7f3cc7e7a failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"122a7f8d4d41e76b09582b047aea8cfab6432530","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 19:44:33.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":7},{"id":"9a8a052321b4134f11d9671f9339cfae0abbbc05","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 20:38:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939572 : FAILURE in 1h 45m 10s","accounts_in_message":[],"_revision_number":7},{"id":"8ee998235dfedc9b38cf1257f4fb9acc01eda756","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 21:08:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/961818da782a4392a696860590717b25 : SUCCESS in 59m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bcbdc94f75014e989b612da0af27298a : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":7},{"id":"58604ada3b0f1cfa0c0e724bb052779d3b3616db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 21:11:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fb3861add6a4a4981846a369b965226\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60caa706b3a48aaa4b71a4a08f39729 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86aea010d95a4dabac585aeb6b039798 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/596279cfb1a7499db124a5bb8f16fad3 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8afaaf367945b588e5d903bf7b96b4 : SUCCESS in 14m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1af267cca0534247962c1c8a9c30a0b4 : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/d95f4b6e75384cb88cdcb33af0350c85 : SUCCESS in 34m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f481d05e2a01420eaf3b302cc534dacf : SUCCESS in 33m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ddb0314b88a4a9f91ca846f8df80661 : SUCCESS in 1h 33m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f7ad1b7688e415c8c62f7e22fd9b031 : POST_FAILURE in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c4660bd291486f805dedd70044930a : SUCCESS in 13m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e8a28cdbbd04ce3aeba234fa7ebe6ea : SUCCESS in 21m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f503d00811d425890e1922927f8abe7 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f0ad8704ca14fa2b0114ae920f50d07 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6079caad18b64deea557d0bb1bc9f608 : SUCCESS in 4m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa387766bb1c4ba88aa1ce4f3c8c5151 : SUCCESS in 10m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be3981895fa247b0a0a454f17e9699c9 : FAILURE in 22m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdb338de50c0495e96dd97a1e187008c : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d637448143140e4b8c4bf862b4891dc : SUCCESS in 1h 42m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2684cc09fcfa4bc8a2b579d42da7a54f : FAILURE in 1h 56m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dffc547de5a4ba0bf50b720470d73ac : SUCCESS in 2h 09m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0c4080c12e9448f7a4b43e223d6263a9 : SUCCESS in 1h 59m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d698850920b642af9a0d1042862eff45 : SUCCESS in 34m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/96a17971dd34426189211019eb7b1e9d : SUCCESS in 50m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/89ae8cec00c24c2d9f7074ebf952d54d : FAILURE in 24m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/107eee0e59e14529b367815624baca2d : SUCCESS in 26m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/611eb5d3f4e149dab5eff02d13683a46 : SUCCESS in 1h 11m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/c81132d3444b46ecb4bdeb35f327a22e : SUCCESS in 2h 16m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/038795d01b92426ea74490b4f852cdac : FAILURE in 12m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9247f1e3776b4afd86bb97c7932dfaa0 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":7},{"id":"cbac41e3db89af8adbe3c7d869b7c41881edd6ef","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-10 13:58:24.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"be7605bc11a08ddccf81a42bed0a7e942ce861df","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-10 13:58:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"65b1863e140667c756de34ea072c1bdfca1d97f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 14:42:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa2600779b8491ba10c0ca47ef578dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3cf1324f05324618a1528f99e6aa9359 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7e1fbf6f2ac426aaf517b387625f2c3 : SUCCESS in 23m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d14c6270874744138bf91ea5165f6ab5 : RETRY_LIMIT in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"622075def7f487c0f598c37bf994cca486d712dd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-10 14:43:26.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5d0e36b82b2d9811d8a3b0144a37604ff11167b6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-10 16:21:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3374b9fc3c5b4acd98e4264fd81d3439 : FAILURE in 1h 05m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d2a5ac172eaa437eb85e407674bd61c6 : FAILURE in 1h 15m 09s","accounts_in_message":[],"_revision_number":7},{"id":"95c40fa5167526dc9de011e247f5cff1196ca777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 16:52:16.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/072384edb2224ac8b3bacb8a414815a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57f3bb18a254a8a8119aae068038eee : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/316cecb0554549dbbd4421a4435209e0 : SUCCESS in 12m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79e1c2deb58b46c4992b0831a6ee043b : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b25aac00f964309a68b4e765ccfc8b4 : SUCCESS in 7m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a0f4763de0b41d8b5adcec77fc8bc8c : SUCCESS in 2m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/0d0b26c42c2f4e4185a2335d60cc0cc9 : FAILURE in 6m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3599f71acbd445c9be2c6bc80c0953d : FAILURE in 13m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/966d3b5c51f6467eb81bc1f9799eb989 : SUCCESS in 1h 09m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f58615ac56f349459ffa03b472819c7f : FAILURE in 49m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71fe1ff6ef544a7c8e4bc18e9f803f80 : SUCCESS in 13m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26a1d260a66c4112abf5c6e76f16fc04 : FAILURE in 27m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ddc2aeb06974136abe332de322c5679 : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dde2c274b3ee42a3b1ed4112841fa92d : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3b05048fd744db295d2b08ecb0bf4b5 : SUCCESS in 6m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59907b221601458ebff6befa4d036bf3 : SUCCESS in 3m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fdf827358d5d4bd796cb5c66bbcedabd : FAILURE in 21m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f722e2b4ff74ac3aaf474c64e3e45bb : SUCCESS in 3m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/973f556fda4b4786b16297817cced194 : SUCCESS in 1h 55m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/311b65a415f24e4dabf270ebbec51297 : SUCCESS in 2h 47m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a82162bca9614bc0aa0e27f54c7c1193 : SUCCESS in 1h 48m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/30a5d6bd4a8b4e2f8d2750ad30242474 : SUCCESS in 1h 28m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3a6aa1c155cf4c8f9bb14fde039ddccb : SUCCESS in 38m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5fff7d297b184653a028494cf9a646ff : FAILURE in 16m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc49bc82312944c3b17e48d90f164e51 : FAILURE in 16m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/11852acab6454944ad102cd642eb66cc : SUCCESS in 41m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16cbce37ae1642bf965740d66e5b7f50 : SUCCESS in 1h 07m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3fb00fcb98fb4342b56540afc2220944 : FAILURE in 2h 31m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c857de5af6ac453eb7f2159c99a9f1a8 : FAILURE in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d4ecabf0ca44d1795815875e4121f76 : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":7},{"id":"8d5d2eb691e6392ccb6a37a4c69195032c003044","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-10 17:40:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"11c398a8961de8b60a37ec602a4bfc36641f57ab","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-11 03:50:04.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2ebef830118e1d0f4b02f2c55dc626886eb86765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-13 16:13:31.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":8},{"id":"ab60b98e4ff9bad17bbee03011405c58074fda0d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 16:48:18.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":8},{"id":"f0c9491a08d45b6857beb9cec6dd3cd133620e75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 17:06:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0100666443f4d9f943f108b143b748c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0421ba0f69e64a789434aa355546dde9 : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7924cddd1ec4efa8b24096d66d81cc8 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb0d9f2726844d5292b95220f14805ea : RETRY_LIMIT in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dfb0c8c7addbd9a30827e2230c9fefc35dbe879a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-13 17:57:36.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cd80a53036c975c667f2cf8e51b160c8fbe0ae06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 18:33:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76894ee726c24d6e8a048378c8890805\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7545d65d264f61b212ccde89e19636 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7762e3e39e2a409f99f01ef9076a02ff : SUCCESS in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4393a389fe9f44338527c7ea6d2cadcf : TIMED_OUT in 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320b775decba41cca962d3c1417e44ea : SUCCESS in 13m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d96eac30f8f4c90a2a94a780a5b54af : SUCCESS in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c20882a85c294114abb41218fbf2a45e : SUCCESS in 30m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b6309e7e93e46b98df609092c63d41a : SUCCESS in 1h 06m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ed8ccc905064137930aa15373cc51d0 : SUCCESS in 1h 47m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81b2426eb79c4f5e9716427345b86855 : SUCCESS in 1h 05m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6d4a7a17a9e4ec790d316bd0797246d : SUCCESS in 19m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2c4b51bfff8f4ba88727c32517143d4f : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ea10b14ae240498487ba932556642221 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fcd53e73b14f4582a9dc792f8109e8dc : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6ad954bde5542719bf2f8b38e76daff : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/44185a0b4882465d966eb7bf6363d230 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fc1c22251a854fb592d9bfdb819cc588 : FAILURE in 37m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8e33d4363be4c3ca1dec897475c8afe : SUCCESS in 4m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20645dde701f4367a8c35e8629ab91e4 : SUCCESS in 1h 56m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ddb90decd1348d09a4502e79e6aaca7 : SUCCESS in 1h 24m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9306d0da74d4d89b7fa39139ffa4a82 : SUCCESS in 1h 43m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9dcab19a7aae42edb5730f48f74acbeb : SUCCESS in 2h 06m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4f682979ad7e4094bee5ebad587dd6e8 : SUCCESS in 38m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d0a6d0861aa543f3b6555175c5efb5cf : SUCCESS in 53m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/573e124b41d3409086d3c1a75ec39dc8 : FAILURE in 25m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4a1b4a504694838b4895d594a1781da : SUCCESS in 48m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/325a01dbc876494eb5a384232312e17c : SUCCESS in 1h 09m 10s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1bcc916a4233484eac76ded59f6f5e79 : SUCCESS in 1h 08m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b70f66731a2f4d0aa1ec8cdb0b3043e5 : FAILURE in 28m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/036da29ea9124dcabda47cc4ee6b95ae : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":8},{"id":"86e7625cad40a6c590b1bff790333250fc29d9ef","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-13 20:37:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f87626af4539e6b293b99fb88755c2005dc8758e","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 20:43:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d4ddd48e6828489b88221be095a59495 : FAILURE in 23m 03s","accounts_in_message":[],"_revision_number":8},{"id":"5709877348ae82d043f554a95355e2916117cae2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 20:57:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b660d1ddb3f471aa20cd50acaf7c7a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5875b3225f7b4314a1e42b6c2a50667a : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f54547d2e2f44ef5b68134b1bebdf03f : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8e693aeb896a4108be00816246751072 : NODE_FAILURE Node(set) request 188bf49b3929426daf3763479e2a6d7c failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"327f029e0fa75b4491b8ecba62169bf5eef7349c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 23:24:24.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62781e7f6b1f4cffb8b0e8532d4a4503\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d35d6bd7a55406783323a7dfa92a4f6 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b68c0bc01c744e8ab6b06c038ffb6b39 : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/525bf6c598f1433cb4940896509ac99c : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf44b5e4d77f44c7b297ebb6e10f8733 : SUCCESS in 12m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03df6657883e4ee5ab41841c3a87e1eb : SUCCESS in 2m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3611745cd7204f20a2a3e5027049767e : SUCCESS in 50m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a50c134fcc0240fb93cb0fab815a6321 : SUCCESS in 1h 08m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c3856157f694d1d8fdcac0f12fc96b2 : SUCCESS in 1h 29m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f21ec7f3be4942a0ae3a7f1c8e0a00f1 : SUCCESS in 36m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bac1ff952e634ff1a6f0550b332ac422 : SUCCESS in 10m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ce3bec0fa7a405c89db46a866de2b63 : SUCCESS in 21m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4b03c02445d14f319ea0df4762c1c022 : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ab05e56b2564fe2b80dcceebecb33a8 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ba0095423564d69a31bd310437ad237 : SUCCESS in 3m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/50a52aabc851431aafcfd1d685a0d95e : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c30302cd4f444ad6ad82a8507b369d30 : FAILURE in 26m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0f51e21893443ae98678053089598f5 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c06ea3968e0495b8a870e150ac1e432 : SUCCESS in 1h 49m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf1516452d594027bcf276affcbf0c33 : FAILURE in 2h 32m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cbefdc25c630406fa17710c40d8ee543 : SUCCESS in 1h 39m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2e3ca0f9b27c406e9f7747ef559deb4e : SUCCESS in 2h 39m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/798c7e9d7ece45f7a9dd6b43205ccc4e : SUCCESS in 19m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b47a145367a441b88ceec29e32d3a69 : SUCCESS in 1h 14m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44466769474e48d7aaa54e28d61ec25e : FAILURE in 28m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5873e89d6955457ebd3f34ef282c6e74 : SUCCESS in 41m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c64f9725ced49c9a4b0410aac04c618 : SUCCESS in 1h 56m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/83843be2d32e471a9c26767368c930f6 : SUCCESS in 1h 29m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfca8c353e0e4ebab7d474c2d5c31208 : FAILURE in 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7cdaa99810b479c85aee9e26e80af9d : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":8},{"id":"a939ea52bd5f8d28dce6ec84a1bb1e3ded51daa9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-14 02:47:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939572 : FAILURE in 35m 05s","accounts_in_message":[],"_revision_number":8},{"id":"f3f7f83cd6e304f70a0395356de5c067b4f7085e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-14 03:31:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/350bbe1167b64c86975a4dcf13e1af0a : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":8},{"id":"d2e444c1afc1ae86df06a5c12e0b3ad56358eb5a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-15 03:45:33.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_664/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":8},{"id":"d4f3dc8c00e2a539a5406f796f5693e169e0a581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-15 04:30:08.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"8da5c9ea6c2be53aa5782487fd45ec54d033c47c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-15 04:31:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9c991e4045a39e25f5abaf31987b60f672634062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-15 04:39:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7482d43a039a8f7f6f54a126791059ba027d0188","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 04:55:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4328a7f13b7747f082e4336798c352d2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d76a54847e65482daefac0dd176c09cb : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3dde296ecd1437fb4e6c4e2513f5cc8 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2bd27cdb316435dac72108fc1cd3a09 : RETRY_LIMIT in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"afa6a8658c662f2f60aa5929eef5e031ee48748c","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-15 05:04:54.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c801a13957194be787ea059ac54c18e8 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":10},{"id":"874eb997578c1838350fefb8afdef28647f6b480","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-15 05:05:41.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":9},{"id":"82efcc8ce7579ee8af68556e7a93ea8b06673bda","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-15 05:14:17.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885 .","accounts_in_message":[],"_revision_number":10},{"id":"4404799d31a20dfa3641e2080ed830e51bd62f83","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-15 05:18:08.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/939572 : FAILURE in 34m 31s","accounts_in_message":[],"_revision_number":10},{"id":"fc80360fdddb86443a78dcbbd54aa468ce5e641b","tag":"autogenerated:zuul:check","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-15 05:22:11.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-3par-fc-driver build/485435e372b843d3855e961f9ec04e9b : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":10},{"id":"1505859cebf7feffbcc38cebdca8e4e1e95a5d49","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-15 05:24:09.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/25d43c39d9ce40519b49010ee5a53d61 : FAILURE in 39m 19s","accounts_in_message":[],"_revision_number":10},{"id":"4683a481b9716572bb8054203efa4465cf29a475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 07:30:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59386b7cc10c415e9bc76f4327d99c8b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f5d646ceebff4136895442e3cf11e1d4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e14465c8123445cb36c95e2b81dcc8a : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dcaffe758e24833981a84aaf8d5cfdc : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfdf0cbb636d48439321438e434bd578 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58448d243c14567998b08e231b665bc : SUCCESS in 10m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc108fb6a1954c75b7cf5e58a6865d40 : SUCCESS in 2m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8909c32d7544406895e354ca24943905 : SUCCESS in 41m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2684b4e7893f4412aed0f8f8b4d7c04a : SUCCESS in 49m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c0902072712410082e5293fa4a6335b : SUCCESS in 2h 04m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3afde4e3071423eb6a37498fdc3b9c6 : SUCCESS in 52m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/452017e1cbc34039af51deeb85e3cdd3 : SUCCESS in 16m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c6438727c7244ac815b99624036b276 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2afe6161061a47b28dc9e778f81e22a5 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76b4b4bb103c4658a3ae2d837c3b0a7d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07bad9bd749047df9cef6b97f3719cd7 : SUCCESS in 6m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8b4854d142d44fc3988984011a49dd9d : SUCCESS in 8m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d61703120cdf4a1880bf91cad951fa3d : FAILURE in 41m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/308ca61dba7b4cdba23e93c311fd0751 : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/111ec2fa57494644841d42ca9608e99e : SUCCESS in 2h 01m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8a42500ee91245c7b1c75696a407a118 : FAILURE in 2h 39m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71c3b4b1ad224553946616e1aae4afbf : SUCCESS in 1h 42m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/665db283af07459098c6637678e0ce2e : SUCCESS in 1h 52m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2fc29453e3a84aef8f11bdd5a92ab755 : SUCCESS in 18m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/56b369cea1a7436780e3f31048be25d1 : SUCCESS in 1h 19m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/367b4e11ac3246d29a5fc6c7b9bae413 : SUCCESS in 1h 57m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ddbe30c3790643b68fad99e0ff8edfe6 : SUCCESS in 41m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6dfe8a0ae4d34c87a8642defb1bdbabc : SUCCESS in 1h 09m 05s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/fa8c9c112dc24e479c4d7d916b08cde3 : SUCCESS in 2h 02m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bff9446685cc4b41ab9d0cfcbd241772 : SUCCESS in 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea94bac8380d46d3b677a359d8c176c1 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":10},{"id":"459daa3d294a38a4f3a93327e33da836e0b22168","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-15 07:40:41.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_02_2026_07_33_37\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"9c21ba7b585ff996c027bd3cea0fb48bdf73e3a8","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-15 13:12:52.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_670/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":9},{"id":"a0cb4e0aa25945f267b829fc1aab4944412416a2","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-15 14:40:19.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_671/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":10},{"id":"a1a7c2e7f3a1c2b9f702612d44948df9e07bb678","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-15 22:50:21.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d14dd68a3a894d5a9c18a454d995ef32 : FAILURE in 51m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e9e3c37a6aa840e9878f7cdbb9c96fa6 : FAILURE in 59m 30s","accounts_in_message":[],"_revision_number":10},{"id":"043fa8d27e4bbd26609f23a994e19272f9e094f3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-16 14:31:18.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"74264935ae09769fc130809b2b5e8e340a57b399","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-16 14:32:49.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"e049f39695384f219552191d828c5d4ce87674c8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-16 19:51:33.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"fdab37716af6d6de55f41f7a18a7f632853e46f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:51:57.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"554f117bf3dd1ff9eb655db97528e63b45fe486d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 21:31:34.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade64dd786e143c78cd5f93789814ec1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d5caf088e444c5dad5db9b84a25aba2 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13682a272e9441ff9b48240a9ed5a2a6 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a0806a01335468aa92559b1cd470496 : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/840fd3108052471384c381c21e0a6566 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f0f7b8e96748fb82f5d76005623dbb : SUCCESS in 15m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f1d60ae6b004493b3b84fc0023db476 : SUCCESS in 3m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/18fd1d57198044f4af266906cad3b1d4 : SUCCESS in 34m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/458ae76df9fd450c857d0f4bf9b5593c : SUCCESS in 58m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3af6613d92ee4100bc47586e69689600 : SUCCESS in 1h 30m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/157667618af0424f9c409271b6410f3f : SUCCESS in 33m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51c5d8754c514970928744468f57079e : SUCCESS in 11m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ea87f44f2594d28a36b74cfc4525bfb : SUCCESS in 1h 10m 50s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd8407eb3fa841ca8d3949ce6dc835e1 : SUCCESS in 1h 00m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/384355333d844e99a74a95fac18a3334 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":10},{"id":"c5a072a37dd626a9471c2e16469ee2f37e722f08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 21:31:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2bd5ccf717b4fa819e274eb38c7792545f302106","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 21:32:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36242f7e34624db697e52e4d45ff6a3c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/147691abf8b24aad986135a0c642dc3a : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f57b6d2420481eb6fbd329f1125a44 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/126f34d7c9f64a2083d4ca371110c965 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":10},{"id":"049826967b4f932d278590785fa4091674dfd731","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-16 23:12:41.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_686/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":10},{"id":"85b9ad6fc3d778f79dc5011f77786cf79d40dc65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-16 23:12:45.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/686/ : SUCCESS","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","revisions":{"382d6effe0fa13137bad73ef6dca6c1ce679fe69":{"kind":"REWORK","_number":1,"created":"2025-01-18 05:51:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/939572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/1"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:49:27.000000000","tz":540},"subject":"wip: Remove Windows","message":"wip: Remove Windows\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/382d6effe0fa13137bad73ef6dca6c1ce679fe69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/382d6effe0fa13137bad73ef6dca6c1ce679fe69"}]},"branch":"refs/heads/master"},"bcc26c9bf9006926b4414478bcf883822a8fd6d4":{"kind":"REWORK","_number":2,"created":"2025-02-21 14:26:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/939572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/2"}}},"commit":{"parents":[{"commit":"5414bec1ad6abe363a7614fb12932a8989258328","subject":"Merge \"lightbits: doc\u0027s \"retype\" support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5414bec1ad6abe363a7614fb12932a8989258328"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-21 14:26:12.000000000","tz":540},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcc26c9bf9006926b4414478bcf883822a8fd6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcc26c9bf9006926b4414478bcf883822a8fd6d4"}]},"branch":"refs/heads/master"},"0db07b76763fbd83fdf74a46d65bd0c9d34936b0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-20 16:31:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/939572/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/3"}}},"commit":{"parents":[{"commit":"5414bec1ad6abe363a7614fb12932a8989258328","subject":"Merge \"lightbits: doc\u0027s \"retype\" support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5414bec1ad6abe363a7614fb12932a8989258328"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-20 16:31:55.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0db07b76763fbd83fdf74a46d65bd0c9d34936b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0db07b76763fbd83fdf74a46d65bd0c9d34936b0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"57c6c05a19010bfdc32a53a39e4ea4bbbfb436c7":{"kind":"REWORK","_number":4,"created":"2025-11-20 16:32:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/939572/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/4"}}},"commit":{"parents":[{"commit":"5414bec1ad6abe363a7614fb12932a8989258328","subject":"Merge \"lightbits: doc\u0027s \"retype\" support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5414bec1ad6abe363a7614fb12932a8989258328"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-20 16:32:26.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57c6c05a19010bfdc32a53a39e4ea4bbbfb436c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57c6c05a19010bfdc32a53a39e4ea4bbbfb436c7"}]},"branch":"refs/heads/master"},"8810230b1deaafcb3bd7a16e38b1e98d56ca3844":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-20 16:32:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/939572/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/5"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-20 16:32:34.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8810230b1deaafcb3bd7a16e38b1e98d56ca3844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8810230b1deaafcb3bd7a16e38b1e98d56ca3844"}]},"branch":"refs/heads/master","description":"Rebase"},"23d1fa8487393dae0192dccbed74cc4e8dfe710c":{"kind":"REWORK","_number":6,"created":"2025-11-20 16:32:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/939572/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/6"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-20 16:32:40.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23d1fa8487393dae0192dccbed74cc4e8dfe710c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23d1fa8487393dae0192dccbed74cc4e8dfe710c"}]},"branch":"refs/heads/master"},"ec029fb1d49741986687afd85cbda88f1a18838d":{"kind":"REWORK","_number":7,"created":"2025-11-20 18:48:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/939572/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/7"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-20 18:48:47.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec029fb1d49741986687afd85cbda88f1a18838d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec029fb1d49741986687afd85cbda88f1a18838d"}]},"branch":"refs/heads/master"},"2db0584eb8a2abc12a7c59616dfa5d86c434b9c9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-02-13 16:13:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/939572/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/8"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 16:13:31.000000000","tz":0},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2db0584eb8a2abc12a7c59616dfa5d86c434b9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2db0584eb8a2abc12a7c59616dfa5d86c434b9c9"}]},"branch":"refs/heads/master","description":"Rebase"},"97c69173d2fc4e6fc078b9f3ac9a1b5cba82c19a":{"kind":"REWORK","_number":9,"created":"2026-02-15 04:30:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/939572/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/9"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 04:29:45.000000000","tz":540},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nos-win depends on pkg_resources which was removed from setuptools\nrecently.\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97c69173d2fc4e6fc078b9f3ac9a1b5cba82c19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97c69173d2fc4e6fc078b9f3ac9a1b5cba82c19a"}]},"branch":"refs/heads/master"},"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49":{"kind":"REWORK","_number":10,"created":"2026-02-15 04:39:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/939572/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/939572/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/939572/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/939572/10"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-18 05:32:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 04:38:57.000000000","tz":540},"subject":"Remove Windows drivers","message":"Remove Windows drivers\n\nThese drivers depend on os-win, which was already retired due to\nretirement of WinStackers project.\n\nos-win depends on pkg_resources which was removed from setuptools\nrecently.\n\nAlso remove flake8-logging-format, because it currently requires\npkg_resources[1].\n\n[1] https://github.com/globality-corp/flake8-logging-format/pull/78\n\nChange-Id: I6bdf6e284e427cd6415cfe40dc1a75e9fe4a1885\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
