)]}'
{"id":"openstack%2Fcinder~940533","triplet_id":"openstack%2Fcinder~master~Ideae7b290c639ae1a79d29eeaf156de38777ae8d","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-02-05 21:55:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideae7b290c639ae1a79d29eeaf156de38777ae8d","subject":"Rally: Fix usage of cinderv3 service name","status":"MERGED","created":"2025-01-31 15:24:02.000000000","updated":"2025-02-05 21:56:53.000000000","submitted":"2025-02-05 21:55:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940533","meta_rev_id":"73840aaf128e5d06c24ad2308dcff754aaf9f154","_number":940533,"virtual_id_number":940533,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-05 21:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-02-03 19:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-02-04 14:53:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 16:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 16:07:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 16:17:52.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-01-31 16:33:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-01-31 16:47:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 16:59:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-31 17:04:44.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-31 19:08:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-01-31 19:28:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-31 19:30:10.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-01-31 19:30:17.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-01-31 19:30:57.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-01-31 19:31:07.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-01-31 19:33:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-01-31 22:00:20.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-01 04:31:44.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-02-03 06:53:17.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-02-03 19:21:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-02-04 00:47:12.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"}],"messages":[{"id":"0d10585d5c308be078d12896f348054baadced56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-01-31 15:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c018911d897c8bbe0e24107f3c69ae65818a04f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:07:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb6165282f21462392698c08772595d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02e10d6dcd3b4c2abdb7ff15c9eabe03 : FAILURE in 21m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/924d3b21be104195bccb7639ec5409cb : FAILURE in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6b5458185ebfd38b1d7bffbe4ccf824048c45a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-01-31 16:07:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/940533 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":1},{"id":"668fd97e0b92406c1e19a9ecafde5d572409a4cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-31 16:17:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/322/.","accounts_in_message":[],"_revision_number":1},{"id":"3bcbc40394dcdde023d0a34235a989cba50fa9f8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-01-31 16:33:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/940533 : FAILURE in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9145573d1786c57bea8d67378e5753481655d861","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-31 16:47:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/67865c3997a14829b940859a4c7fb996 : SUCCESS in 1h 18m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/277d3b15f3a049e48ad1b6a1b60ea876 : FAILURE in 1h 21m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4602e575f9376510ca391aa05ad0376fbb86d472","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-01-31 16:59:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cbb8d2e12f8746b58b3a9fdaefa0866b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3faef811a63c481a9aa8e01088f7e1fe : SUCCESS in 44m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe89e985c5c2487fa1cbe118d159f7c0 : SUCCESS in 42m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e411e43ab46541869672e1e34a560492 : FAILURE in 1h 34m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb940d05f47b47c4a2c878d25f4d9063 : SUCCESS in 44m 26s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d271437ca9104be6bc06258b7d2d02dc : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1bed4bfd9b9c4fe78d376a07edd561e8 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5b7661b0688d4c87bf2d2bbb2b8b2597 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb24114bbc3a413fbc25043c2ed47a92 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"b022e6b4098fd162c2a6ed7de5d15304d7e24dd6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-01-31 17:04:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/33/940533/1/ : FAILURE in 1h 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9f9926c6b7c753864abab231308180fd44d42ede","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-01-31 17:53:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/303/.","accounts_in_message":[],"_revision_number":1},{"id":"e7d63e2e8db07aa3186404bebf3194e619c96d48","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-01-31 19:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-940533-1/FC/ : SUCCESS in 3h 43m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-940533-1/iSCSI/ : SUCCESS in 3h 40m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9d994d245c30292dd167746a17b5621b91b8791f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 19:28:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3157dfe45f534149a6e241ec96f165ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c65f2d63a1478aa4df5597a4de3cae : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07649c10cb3d437ab9f630b4b997f7e9 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a18c8dd15e547d38aa06d1c6bf2680c : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273afd8532d049a895481f7dfeaa1e8a : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0c759b6468364f7f88614db00cd8b9ee : SUCCESS in 1h 29m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88ba7c0521014c4b883936bebb330fa9 : SUCCESS in 57m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db69acc25d294ef4a1f3684a9079b4f0 : SUCCESS in 1h 35m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a56228a5375d41fc943cc0a43d925bc7 : SUCCESS in 1h 06m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/73152da8348d45f38b2416f59ce91d85 : FAILURE in 25m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f795e4cd6be4a4aa2a945956366d11f : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a3aac5bc09642c7953d8b00ebf47a09 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2fc49cf9bd514277a958f0e2a4655d0f : SUCCESS in 7m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/262d488b4bc3455fa2ceb05fb38d6c48 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/40655de1f7d24125bb28ced724a83055 : FAILURE in 35m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd9f5ade0f8642629e7fd92625a72df3 : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b266faf0b17e47688097fbdcf971d06b : SUCCESS in 1h 55m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e149c1efe06e48d3968e4b53f7433946 : SUCCESS in 2h 38m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01e22559f6f242458b424f2f66bf0a73 : SUCCESS in 1h 41m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7c0f52f912fb4be291d2e1cd38d03830 : SUCCESS in 2h 00m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db76b0b92f5947c9b19e73eb8ff2bf48 : SUCCESS in 32m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e9ad2bfa5bc4bafa1eda05711f75a4a : SUCCESS in 1h 23m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/963449fddd18433097db225be4c05eaa : SUCCESS in 1h 53m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a66f070832224e609bb8495a9939363d : SUCCESS in 40m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3a7270c56f0d4c58b3729a8a385032b9 : SUCCESS in 1h 44m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/94208095bc6d4dea9d5e277269bf6ed7 : TIMED_OUT in 3h 34m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb55579500af410c99f0815f00934316 : SUCCESS in 1h 34m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ea3dad225fc4930a2194ea4872e8739 : SUCCESS in 1h 57m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b47cfb63e054c5e9542b97583c05bf0 : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bdf1ae376afc89197b9a10e8089ed6a71da3f825","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-01-31 19:30:10.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"b6c8b1352315870a4c77ca37ca7b39d1f8762274","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-01-31 19:30:17.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"ab22d068fbf6e935a29a84d63e94e5f0d8161a96","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-01-31 19:30:57.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"0c3d9837b43332398450c36eba09f19709241925","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-01-31 19:31:07.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"c45273830f560741ab11030bb6ebf89ee8234eb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-31 19:33:57.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2025_19_29_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a7f871f7bd0e4fb7d434b59d94296d8bd74e7be","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-01-31 22:00:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/33/940533/1/20250201010739/NEC-ISCSI : SUCCESS in 5h 26m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/33/940533/1/20250201010739/NEC-FC : SUCCESS in 5h 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"87cba08be1cf9f8f3b8385c696e0a22ca9d10b63","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-01 04:31:44.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/33/940533/1/20250201044057/NEC-V-ISCSI : SUCCESS in 5h 20m 29s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/33/940533/1/20250201044057/NEC-V-FC : FAILURE in 8h 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"da20261101bf11f5f6e4892bf58189d8941fa0f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:17.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1288/","accounts_in_message":[],"_revision_number":1},{"id":"cfd45f712bdd0ecaebb18c68ec8c8a2387923623","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-03 06:53:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/940533/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/940533/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"3704d3c81778dbb951263823fbf96c5de526f735","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-02-03 16:06:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97f5afe6c43aa945716616ee086d8f58d251089a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-03 19:21:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9d0beea31a78dc4bd162cb4b01467692b286f07","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-04 00:47:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/940533 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c34606e48103d7d7bb96c15244b661cc86356527","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-04 14:53:16.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"723887215b62e57402af1f63f1958e89716676e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 14:53:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51c4a8f35732605a1c9f9ca9fb4c843f110e435f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 16:56:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23b85c27e9c14196a1d1d6843eeb705d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf401a161571446b8b4b36d9509070f3 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e93d6ef0f37d467e87d683945d63eae3 : SUCCESS in 9m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8fe87abe14944abb21f26d5a23273a6 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31538246160947fca9d2ddf2dce47711 : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/b25ef2807da0433a96833809fe386e59 : SUCCESS in 54m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ffaa22725314d4bab02c123235b8535 : SUCCESS in 36m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b96f26703c3c43b9933c7aa2cd8df9ec : SUCCESS in 1h 26m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5f730ea7e004cb99bfbe08ed7a47e2e : SUCCESS in 1h 05m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4cc1f77edc65474990263b3a98cdbcde : SUCCESS in 1h 06m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab58db33c3c547b8aded6109db3ebf56 : FAILURE in 1h 14m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faddce0cdf914840bdf0c99699a7e0c5 : SUCCESS in 1h 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9dccb76cd51401c903808cf21606b85 : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4bde1a073ebe927b408a717280ef242781b34c9b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-02-05 12:58:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c69b0e32e47d42046c8cdbb104e9e8343256d3ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 13:18:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bea1073758fb4b9ba43da68fe1aaa872\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5093f843dfc5418298a0e1ff24f3b082 : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19b148aff36e4104a36f7d7382392e61 : SUCCESS in 18m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f815f12715de5b744e5e0ee45736972b4e6f5b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 15:47:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7f242b4d3df45c6b909b5dcb8e1662d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc59ec0063b4d06a48b662686359b00 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be9080d226974a45a868cb138d86f049 : SUCCESS in 11m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77b8c8828adf42e5a6ac182ca509b0d5 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f324784984594656953d58cda342731d : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f0b58a99eb89423698cf3e077998cb97 : SUCCESS in 1h 03m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d5e7f698c3040959f0d3fe7416903f2 : SUCCESS in 57m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6accdf21d705484490d63ea8e83a6acb : SUCCESS in 1h 41m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbcee60ca1c445cdb96161fb0b2ad38e : SUCCESS in 54m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3becd83b4a0042f49aafb1c37ea5de7f : SUCCESS in 23m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5ae7b6db1434f258fc13d3d4ae9f2be : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30662ed9b18c4d47ace11107c828b73f : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/415554a8243b467ca8d86cee6f41d75a : SUCCESS in 5m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8a42766cc08149498a41ff0937f7d24c : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1d51abcbf2148e2bab98b11b77adecc : FAILURE in 31m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1b267ec032d406d9c0c34c1dda12132 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/312379b08c184940b2a1e20d0a2acfef : SUCCESS in 2h 02m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce7d28c52bb24c5ab1977447f671057e : FAILURE in 2h 41m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1714e7a07b7c499a87ff1ff0b3e8fe77 : SUCCESS in 2h 05m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9644c2b2993045d786c3e4bd11fb27db : SUCCESS in 2h 15m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/123b4dc288ab451e8ceaf11c795abbd6 : SUCCESS in 41m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9276d13679e4771a05ffea620522bb2 : SUCCESS in 52m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1707322a78342d19963beefce332e13 : SUCCESS in 1h 42m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/348d56c0ba9347148a3a65f669b826d2 : SUCCESS in 55m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e7f0117645834f23ad0ab24eeac9097c : POST_FAILURE in 1h 47m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ac3e7dbd52e44b609c19e5338d922696 : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c862ece4872642c38c3b323b6759e0c0 : SUCCESS in 1h 50m 39s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70231b0ea6c04fffa1498b86d3668bc1 : SUCCESS in 1h 37m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac77479a3f2944a7afa71af1851dfa60 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"58d63e9d4589654608862840dc196713f4f71480","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-02-05 17:24:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4107591dd56307867653f8e56439af3b335afaea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 17:45:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2abc7fb38f640ff9a575ceb2a68efc7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c662eee8cd064239bfa083db78267c4a : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4085b229c51441f8ee9fd44977b04c1 : SUCCESS in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec49385c44132698d3dc80864f375bdc8cb63560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc73f5985a44f79ad18d870d0f466fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87cee0b9a458484a9bf1219007ebd528 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/261ca8561ec94b6b9e452e5a0d9e9ac7 : SUCCESS in 9m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da3ef90a1ed641fa9318a0b9e14c3f9e : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4567a57c54b418a866ef78614aaebbf : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/86d9d020446a4268b4d62c9965b1acfe : SUCCESS in 1h 01m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/639595bb9718481ca7c6911e1d0d870e : SUCCESS in 1h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ea5c53698f14904ba8010cdfcc9c62a : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2622f76316eb479a9f72b72e8aa49978 : SUCCESS in 53m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f3a6352e14840c8a1c3fd76603fdf1c : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/255fc23735994447b4df1c11139fd39f : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/869e18b0e5344342ae79b8f9eff3221e : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ecf29b05f72e456a8f61b32bdb2123b4 : SUCCESS in 5m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/31b1e66a3c3b460e9eff208c0a1e14f0 : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/78969bc54c514f15b4fa29d762751618 : FAILURE in 29m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b4cd86491e043efad246d0aba670099 : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a232da77ab24335a91ebb218abbcbd9 : SUCCESS in 1h 44m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a6c069c824b44c1abfa492796b3eebaf : SUCCESS in 1h 58m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe71bef0676645fe8efa8622d1ef8e0b : SUCCESS in 1h 33m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3d50953fafae49189fd331d3553bcfce : SUCCESS in 2h 13m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b3c03de640b34a45ac4f989062dcc444 : SUCCESS in 36m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d270a47e48434442806189793a909f2e : SUCCESS in 59m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9284e187aae4ddc9c60bf9b17a26f82 : SUCCESS in 1h 37m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac2a9b523b274e81b5c7f725b9eb6441 : SUCCESS in 32m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b90eb7b4a35b4409959bbe1e60d860ec : SUCCESS in 1h 09m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5841c82155da451f913ee69a8f211322 : SUCCESS in 1h 49m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ab4967a69624df6b44a3189a45cf3ea : SUCCESS in 1h 47m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4216ace297484230afca6aea24db805c : SUCCESS in 1h 20m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f478eeb891524f49a12340d61fb18ca9 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ffd1ab0acb6df9241c71a34393a42cab52ea3d38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:39:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dca267c7cba5af8e07b99125c25804a323a143e8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-05 19:39:25.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"00f2e3bab025b489096e63af17e8be9a22406745","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-05 19:40:06.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"5d0fb83c1be661732ba3b59da49e708b4639e9b8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-05 19:40:19.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"79f5a3d66fe107298b31568b84974e80c75a7f44","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-05 19:40:51.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"6668b8ae5596eb88d357045e582a63e4d8f8ffec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-05 19:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2025_19_38_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"afbb7e7dd0b96604af58c08f539d4d10909be311","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-05 20:33:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/230c66afc810489c83330969797b2397 : SUCCESS in 1h 19m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7479537c3b22432198ea0760f270432f : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7046a799941b7107547c2c3a186fd6657b9de8d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:55:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd969b909654b19b9f5a5b2c160c40f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffe5e57cfe3462abfde93b57b8b4068 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5df9ca7dc1ea4ce2944977f6dccd70a9 : SUCCESS in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fac7abc885b0492eafa408e2173879d8 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1d36bb6ad34abeafd80689a4c7ee25 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/7a91155e4a0340b58ef9eff70cad0db3 : SUCCESS in 1h 02m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4419dba8da845f8bc0f195b1b248f92 : SUCCESS in 37m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6120c8ec08a4cf3909fea0c88f39bbe : SUCCESS in 1h 48m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d151cb1fa0241639029a02bdeab3a88 : SUCCESS in 1h 07m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7d3c0588d1c47b8876f8bfc206bb652 : SUCCESS in 1h 14m 30s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2213ae2963dc48d58cf884806a834d85 : SUCCESS in 2h 09m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7674f66426614a368ce197f4b675ca55 : SUCCESS in 1h 37m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cec0a11df4541b2a4e7535caae12a27 : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"69843817eacbae24b4e3f220b4669442cd1f7328","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:55:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73840aaf128e5d06c24ad2308dcff754aaf9f154","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:56:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0b27bb2c10409ca044b65db8b48e5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51cbb57f61d434f99bbb0a9a7794286 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fff98b182ebef323f4b38d5bb89db2851a12bc56","revisions":{"fff98b182ebef323f4b38d5bb89db2851a12bc56":{"kind":"REWORK","_number":1,"created":"2025-01-31 15:24:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/33/940533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/940533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/940533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/940533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/940533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/940533/1"}}},"commit":{"parents":[{"commit":"651bb416899007b49bf4de2a9bd76afafb70de31","subject":"Merge \"Dell PowerMax: multi detach req caused race conditions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/651bb416899007b49bf4de2a9bd76afafb70de31"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-01-31 15:19:36.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-01-31 15:19:36.000000000","tz":-300},"subject":"Rally: Fix usage of cinderv3 service name","message":"Rally: Fix usage of cinderv3 service name\n\nThe cinder-rally-task job is failing with:\nValidation error: There is no \u0027cinderv3\u0027 service in your environment\n\nChange-Id: Ideae7b290c639ae1a79d29eeaf156de38777ae8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fff98b182ebef323f4b38d5bb89db2851a12bc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fff98b182ebef323f4b38d5bb89db2851a12bc56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
