)]}'
{"id":"openstack%2Fcinder~941478","triplet_id":"openstack%2Fcinder~stable%2F2023.2~I600881347bed834383cddb9127e24954ed77b05d","project":"openstack/cinder","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-04-10 17:31:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I600881347bed834383cddb9127e24954ed77b05d","subject":"RBD: tpool.Proxy client object","status":"MERGED","created":"2025-02-12 19:39:02.000000000","updated":"2025-04-11 12:30:58.000000000","submitted":"2025-04-10 17:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941478","cherry_pick_of_change":894226,"cherry_pick_of_patch_set":4,"meta_rev_id":"2e3975dad3132e008bb12b71af4c61e26e3eb53b","_number":941478,"virtual_id_number":941478,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-10 17:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-09 14:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-09 14:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 19:39:29.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-02-12 19:46:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-02-12 19:55:08.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-02-12 20:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 20:13:17.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-02-12 20:15:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 20:49:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-02-12 21:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 21:56:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-04-09 14:30:45.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-04-10 08:29:27.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-04-10 13:37:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-04-10 18:44:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"f103e1e8e5a82a54781d11903ba0f558a7d4351b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-02-12 19:39:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d08700392bbad0c06bee57adf5e270cd8143f31f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-12 19:39:29.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1323/","accounts_in_message":[],"_revision_number":1},{"id":"3a06fa88aacf089aa9483f91e6882d160cd91d96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-12 19:39:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/941478/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/941478/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"4f1c368fb300c4050deffda16da2d4a600ae3275","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-12 19:46:38.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ab808ec2fa1e46d4aaf232bc8cf15804\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3d5dbdd3201c4f56afc1625719af2b73 : FAILURE in 6m 49s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f8d536d02304b03818dea21d9d98b63 : FAILURE in 6m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab21e84f05dd48da8e865413dbb9a551 : FAILURE in 6m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3757da859b9548dd9eb091d6c509c328 : FAILURE in 6m 49s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003c... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"574175e1c9ce43086f15b90bf261915c98426a51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-12 19:55:08.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/366/.","accounts_in_message":[],"_revision_number":1},{"id":"0148090f841c491c5110c8491da20c101b40834a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 20:00:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de98664281eb4f1abf51934a562274b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2586dfa569b24031a32aa3a6298c2731 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91a454cd626f4b45b870a9f4f3ab5e17 : SUCCESS in 16m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e91e07861f3246e9bdeff22621a8ffa9 : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4dd2a590aa464d99a62045949bd7f6b9 : SUCCESS in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37214dae594ec253de08e6ffee9c1242f93aa5a9","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-12 20:13:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/941478 : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"699a4edb2e7514ded99553d285e1b5872c9f3ae9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-12 20:15:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/941478 : FAILURE in 34m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b801df5124d49e567499ad2ff0267987ff0988bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-12 20:49:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/941478 : FAILURE in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f5e62b311b2f66b2d920df655ef3f738a61a7962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 21:51:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b557422485ea4d539607070acd1bab55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c920404a051f4e62a09cce5546c9c660 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/495e845135db46b798f655620ab11ba0 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ae3f81756b482b9db5fa4b34349f9e : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00c3d0dcefb64379af9342297512454a : SUCCESS in 9m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8c770006eec4218a95f07fc0bf6b015 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb4efcf08694d2ba52be88170621242 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e792bf29d7c849de8893eeafb21c3dfc : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f375486eb71142fa80a853429aaea1a3 : SUCCESS in 1h 17m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5496817d44441efaff18ce1d1f2fae7 : SUCCESS in 59m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/065e5c71f1bf4c3594813ef5cb429618 : SUCCESS in 15m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/955bb2c020ce4ea0918e06cb41cfae9f : SUCCESS in 7m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f013c1c078b54b2e858a0e8192e691aa : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cd944b9d6774fb88221d0f4c45d41a3 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/386d82ba1c6d441485dcda29d6e70c95 : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e073898174f40829b48d457c0ed7800 : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6189daf270ac4044af984d41d68867f1 : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a711bba11c8049c593f57cb9d0bafcc9 : SUCCESS in 2h 05m 25s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2998ab8b187e413b89040ef8b852993f : SUCCESS in 1h 55m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f013d4d84ce04ea09e673938aeee879e : FAILURE in 1h 49m 28s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4498eabab7a44628b05e433492da6186 : SUCCESS in 1h 01m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/390d06054a784dacaded47c3d3bc556a : SUCCESS in 1h 16m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5dc0e14f1a7b456b980104105620b607 : SUCCESS in 37m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7f5f3fbe44a84618bd8450b6c3159ad3 : SUCCESS in 1h 12m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cb0f28d280a942e1a6811da70f423610 : SUCCESS in 1h 23m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54fad2520efc4f689b3ef2f5c5a11c22 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"419fe9db33d6cd790dde7196794731f2f3e84fff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-12 21:56:22.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/12_02_2025_21_51_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"5d2e8c1021ad50be6c4bb7ac776dcc89755a3b25","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-12 22:56:59.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/348/.","accounts_in_message":[],"_revision_number":1},{"id":"b1bed5ee00deb8ae740c7059570bb7ceb9a053b8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2025-04-09 14:30:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63da9591eab424466b85e8e92b170b29c7c27b21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"531e8490ea71e522c4e769de6c9464efeb4ff396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 16:27:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a5eba817d464f2a91ef334eb6fabe45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac5db92cd36047189961c1d72f021202 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf8ae3d931d4963b5e67da2747a1611 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec32bbd501de45018aa2aa989e17a23d : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ca9edcd6612485f952fd3ee495aede2 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15fe4cf5f05a4d1c8b4bcba15485058d : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/fa84737faa3d42acb957c99c37d8aeec : SUCCESS in 52m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1511bfb991a7401f99b11e9115f46b6c : FAILURE in 1h 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05a138a5b58846bd86160a636dfefd3b : SUCCESS in 45m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7721a56286d24445a1df534c4863a952 : SUCCESS in 1h 03m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40c8b5615dde4987bf318c6291aa629d : SUCCESS in 1h 35m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91ab5219f640406badd6cc783a9d5a93 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f8e0ffd02168b1d3c1189aab90d4e7b6b851268e","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-10 08:29:27.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af1df3df45fe498f71bd914dff7086a483521c83","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-10 13:37:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c697b4bea3abed4053302efec114b8a0bbd10a4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 13:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d370c31aa1794e7289c1769b4aded576\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e05b79f12f241c8816ed62cec076f7a : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d0882d3a4b44fada67a224e0ef1ef9e : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffb2dd1bd0c44a818ff4d190b7c70024 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0b4a38ef60842ecadd4c1f49bde2891 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c94dabf7cea03389bc4b5a6ec872a6dbfda8d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:43:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ea5093cddb94053a9aad183d9322676\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d28a01c771f43819e3e98dca11ae1d0 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fcb3df1bbd44ce892a769fa4c26c79d : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2becb99e51940f982e029eba84d7df3 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d88550ec11194dbea9b014502d646d2e : SUCCESS in 12m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c6f26bcf8404a98b53ccb017e88268f : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109a61d36d554c2e80fd24908b4e8b8b : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/405b1e57d2eb4ef68dd26a7ab6f0be10 : SUCCESS in 42m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8059d65933684d65ac81fcee1f3a5bc6 : SUCCESS in 1h 04m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/820224b5a10944e685dbaf5b0fed991b : SUCCESS in 40m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2306b22dca83442182105fb3ed6e95e7 : SUCCESS in 14m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20227c7259674874940592b28479f885 : SUCCESS in 4m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aeae845cf25b4f9a9605ac4f44c34ae4 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a111da607c54068bef4577117ce63a5 : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eeb6c36424cb4253979799b8baadd9ba : SUCCESS in 4m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/316fcf3b063f45bfa2d79065d754f88a : SUCCESS in 59m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e00748876124e8993917326365084a6 : SUCCESS in 2m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0fbdf89e6b2543998c90b9c680e8b679 : SUCCESS in 1h 56m 24s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1e0240fcdf0495dae60c34d4a8cee09 : SUCCESS in 1h 18m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d67682342c0c433f87a3032ae5b9ebfa : SUCCESS in 2h 01m 15s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8301a58c7c00419c8d493ea2107f5f8c : SUCCESS in 29m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d2a28ebc684b49ae80e5cc65ba391e00 : SUCCESS in 1h 29m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/facef10e50c44a7f8df630163a448274 : SUCCESS in 46m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b6eb32f568b64d0fb5a0f13d7c518c71 : SUCCESS in 1h 19m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/44b06d894d6145f4bf5e481f18a64cb1 : POST_FAILURE in 1h 34m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc9595bfb1e04ee7aa015db174f98214 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"00e91d4ec0a8531111fcf87e3cddfaf371430806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:44:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcabd4b693520aa0d434e803e6b48170eed2b340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 17:31:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e811e14cf3bf478cb0d3364325190af0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac316187bc1417e9b0acc325855dd43 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/566f07797cd34d93ab3856dfebf3d767 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ce61e294e974ffb8709074d53d33d62 : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/993ada797e51413aa3a20f9e01f95534 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3c4616dede430289b2c636929a30dd : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/71efdd7ab00c4096b74ebd5c43665046 : SUCCESS in 49m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd6d4d244dfb404ebd5828f7d5d6a0e9 : SUCCESS in 1h 03m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2694633fc6bc41de94efa204d3622a66 : SUCCESS in 23m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16863bd3a8124a2f9060bb745aed7b74 : SUCCESS in 59m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87d9cfc1601c4e648360b51ae1302659 : SUCCESS in 50m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e47273bfbbbe429f831992bfc646db4e : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8dbcc5a08371112de8a41c21f0efd8d6adab72cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 17:31:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c03ae9e54cd3015cd89508b38197502866265402","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 17:33:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7916a8bd5c7e4441aab6b93466cf7ec2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cba4dede16c476cb3f10a7c28ef097a : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3dc7584856d94f588425c4cd15503f086b1434ce","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-10 18:44:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e3975dad3132e008bb12b71af4c61e26e3eb53b","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-11 12:30:58.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/c9819e9d097847dc8786bfc553804061\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/05bc8830a4de41e5b407ca52e9442c35 : FAILURE in 4h 50m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94d8dd983e890dd7f39eba5b5e3321bfda818ebb","revisions":{"94d8dd983e890dd7f39eba5b5e3321bfda818ebb":{"kind":"REWORK","_number":1,"created":"2025-02-12 19:39:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/78/941478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/941478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/941478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/941478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/941478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/941478/1"}}},"commit":{"parents":[{"commit":"0480073b9ac64f6897fd6bcea72a8d1cc6d474bd","subject":"NFS update volume attachment format during volume snapshot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0480073b9ac64f6897fd6bcea72a8d1cc6d474bd"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-09-07 20:31:36.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-02-12 19:39:02.000000000","tz":0},"subject":"RBD: tpool.Proxy client object","message":"RBD: tpool.Proxy client object\n\nLack of a proxy on the client object could result in\nhangs when connecting to ceph mon is slow.\n\nChange-Id: I600881347bed834383cddb9127e24954ed77b05d\n(cherry picked from commit 0634096238e5ad227b168876e6a7e2ea32746bf3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94d8dd983e890dd7f39eba5b5e3321bfda818ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94d8dd983e890dd7f39eba5b5e3321bfda818ebb"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
