)]}'
{"id":"openstack%2Fcinder~942716","triplet_id":"openstack%2Fcinder~master~Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade","project":"openstack/cinder","branch":"master","topic":"bp/migration-progress-in-volume-details","attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-03-08 09:02:12.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade","subject":"Show migration_progress in volume details","status":"NEW","created":"2025-02-25 17:28:06.000000000","updated":"2025-06-26 11:31:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"13ed3dc8af883e1da55cbeffd02adece6b2f4223","_number":942716,"virtual_id_number":942716,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-08 09:01:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-25 17:28:15.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-02-25 17:28:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-02-25 17:28:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-25 17:28:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-25 17:28:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-25 17:29:24.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-02-25 17:29:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-25 17:34:22.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-02-25 20:12:06.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-02-25 20:21:37.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-02-26 02:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2025-02-26 21:19:53.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-02-26 21:21:57.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-02-26 21:22:32.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-02-26 21:22:46.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-02-26 21:23:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-02-26 21:28:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-02-28 14:34:41.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-05 19:49:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-03-05 19:49:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-03-08 12:02:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-06-26 11:31:11.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"}],"messages":[{"id":"b11a457ec67c516dd99361b1b46339a929cf0398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-25 17:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79a275ff543b616292f8ba103cbbb063e11ae417","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-25 17:28:15.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1366/","accounts_in_message":[],"_revision_number":1},{"id":"09d7e3b79353646a4f24cb9466d72a4bde114e6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-25 17:28:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"bc3f16953d2b0753318597cd4fc8310825d97231","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-25 17:28:22.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/nova for 941476,1","accounts_in_message":[],"_revision_number":1},{"id":"aca03e3800274628c702cac39077c1ffa736e27f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 17:28:26.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 941476,1","accounts_in_message":[],"_revision_number":1},{"id":"cc714c2597fedc348fa18c90e64a2b57662f829e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 17:28:26.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 941476,1","accounts_in_message":[],"_revision_number":1},{"id":"8d47ec700e248c4d4638e1177f234e89d5c6f3c9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-25 17:28:30.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/nova for 941476,1","accounts_in_message":[],"_revision_number":1},{"id":"8fa0ddbf88198014444b9873ce968c0e596ae6cd","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-02-25 17:29:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"645523ebf8813e544c75047dda02757a18d0086d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-25 17:29:31.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"501dd6981641a7897d6d0bc829732387f7249665","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-25 17:34:22.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d2d515159f096a508ca53edc092583aef41c4c5e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-25 20:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-1/FC/ : SUCCESS in 1h 47m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-1/iSCSI/ : SUCCESS in 2h 43m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9adb6a2ba70db65880345b43b6a11f2078495e22","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-25 20:21:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/395/.","accounts_in_message":[],"_revision_number":1},{"id":"03731dfd61258b5d155e94600c612721ded38cbc","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-25 21:30:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b9f8e982c19927bbcdae30d083a40e6dbc2b8f7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-02-26 02:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/942716/1/fc : SUCCESS in 1h 11m 55s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/16/942716/1/iscsi : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3cd231efcb64ca2811065573cdf1ee4de68f1449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-26 14:19:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcd74f3af91130b62ef2348f1d6edc80fd2ce0c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-26 14:19:12.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1370/","accounts_in_message":[],"_revision_number":2},{"id":"c486eaf44afc771121ad390069eb99fb3ecc3a36","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-26 14:19:15.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"1a7985415a99e39b6b4253846e7dde23ab3786af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 14:53:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b4d1bfd60b41b1a075c80f1190ca89\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9edf8b9ddbd340529216baa3f597fe56 : SUCCESS in 24m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45712a7d559e43d498876ac29c3461f8 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"180f511c7206d1b917442d3f5b3d83be1c75f322","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-26 15:05:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/942716 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"eb2de421897e8e27859c01ac1acaabc94a983afa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-26 15:29:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/942716 : FAILURE in 1h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"715cf315af038bfe1da950ba7e9410e4affca875","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-26 15:36:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/399/.","accounts_in_message":[],"_revision_number":2},{"id":"dbf65f9e8f435a2bdbadfeb52f5a4410bb07ef8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-26 15:57:48.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/52480de7fba746f797ee6f907c126d97\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5e8fd0171484e91b435544bfce89cfa : FAILURE in 44m 09s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93264f0a06a5409db794b5191d4a779a : SUCCESS in 40m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfd11008708e40be85abf8e44809bd9f : FAILURE in 1h 38m 10s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f8de7ed704a45c7ab212efd2493adae : SUCCESS in 43m 10s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30ed77f9bb84432995f0fe13962171b4 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf5af4f55cd34ebc9a658e4984f2f9f0 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea5634b5dc254a6d87e8fa5a77d8e8c2 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3989c7b7ac974e0d935444f9af63975f : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack b... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"d7e68cd2501f1bf84a6481682ad782c87cc89c49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-26 16:21:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2b4bc8730e2848aa80889dffc137ec36 : SUCCESS in 1h 16m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dc1a87126f8b4a149ceb6d571b79d089 : FAILURE in 1h 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ff770f3ce43d3c338ab00830b6f39a50ff06e29a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-26 16:39:36.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-2/FC/ : FAILURE in 2h 18m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-2/iSCSI/ : FAILURE in 2h 20m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"49686337f7c1bc527c3232a07deadb658d5b2852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 17:56:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/211690acf2b64414964564e6eb729819\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35b83d098a9046699303ece0c2b48d01 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a901820922d44b5864e80256864a37e : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fcc40ce48564e61a3e814c4a07e8430 : SUCCESS in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b59fe200d2442ebac875e5e20feae8a : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb24eaaa001d4579a155e44daa60fb50 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/129690b62fbb4f5493f85d26c2db6d0d : SUCCESS in 1h 01m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/effaa0ecee1248318d41fd68cb30c242 : SUCCESS in 1h 15m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c091de254ce5450eadcff247e04a3d97 : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af77618ceaef477fa901a0f628fe932c : SUCCESS in 1h 07m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba78d89e92b451da600c60199bc9114 : SUCCESS in 17m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cfa05ae3acb6426cb1c60ba593dcc5c0 : SUCCESS in 26m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/17288699a3774a5c90e5b1d2257110db : FAILURE in 7m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a58e737836a04aed9b9be7ccd733d8b6 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76874841bee943b5837e7709e30eb364 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8844295dc61d4f4abaed5725d1f7c7c0 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c0ca9668db244232b21fe30a4bf146b8 : FAILURE in 42m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf17cf2baff847978e2c92371e47d182 : SUCCESS in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d240eb65fa714e5c97108893c2973605 : SUCCESS in 1h 50m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/adfb0d74703745dcb60af3c2dbd30e01 : FAILURE in 2h 41m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9e713b27dfa468c816bd7032cadfbac : SUCCESS in 1h 53m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6600c6575cb940a5a7da9d3e8a65c8fe : SUCCESS in 2h 08m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1889924df84347359efe7232b16d516a : SUCCESS in 37m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ae5779ce5e94098b0efd0597e9ca9cf : SUCCESS in 1h 22m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd9f0fa107074afdbb819d5a9048846d : SUCCESS in 1h 30m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7945e4ad952245c58fc1dff8a3b04499 : SUCCESS in 54m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/09439c0e345e42e59df83a800b3210a5 : SUCCESS in 1h 24m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b337c306a0fe4d53b342102c549fbe23 : SUCCESS in 1h 55m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a3d454548344f919ab21f47a625f9ef : SUCCESS in 1h 51m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7410c7aa5c534b6ebe3a71aca1c71200 : SUCCESS in 2h 00m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea14fbaca9a54104aa65db4876a7dfd2 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6c450b4fe714a6f20f3b6760e04e80ce5348de08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-26 18:14:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1e6ec7f47f94a8e8482dd5f68bc07b8673ed30d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-26 18:14:43.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1371/","accounts_in_message":[],"_revision_number":3},{"id":"bcdd77ac6c68edb7b1a278ed2ea0df9f72a54e4c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-26 18:14:45.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"405960e87c2222fae8c0e686f15e5ffa120a9a81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:34:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf1d70710ea408982b0580bbb29ce6d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b68ab00fec14c8490435fc3ef65aa13 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f416f06ad0f4ee083b2829ced049dba : SUCCESS in 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a95eb3e1e39d0e54d24fe67155a3f312e4ad3ee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-26 19:00:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/942716 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8d2d1ae6201e345a60ba22a65d1d63c30d855490","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-26 19:23:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/942716 : FAILURE in 1h 05m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ac249121067ef9fa4d7b6ff6029c44616ee2a01a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-26 19:30:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/400/.","accounts_in_message":[],"_revision_number":3},{"id":"c2679bc926ccd0632f0255ebb56bef54867f82de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-26 19:32:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d61a3c14779648758c403eb5e3a5ea36 : SUCCESS in 1h 14m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1e6f1c21a01f4745b3a723d096e5839d : FAILURE in 1h 15m 52s","accounts_in_message":[],"_revision_number":3},{"id":"868eefe57568360f6b523306a4f3332cd146c309","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-02-26 19:33:42.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-3/FC/ : UNSTABLE in 3m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-3/iSCSI/ : FAILURE in 1h 18m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"574b73e3806f1aa951def06e8fcbe1dfc7624941","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-26 20:00:15.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/90cd65070a4a486290b6af2a42d2f81a\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46a2db437eba43e7ac636b3cc75ab197 : FAILURE in 47m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b74a5f842774f2fbd41c423b0eb141e : FAILURE in 1h 45m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/314d02f23e8c40f9bed02adaf99cb993 : FAILURE in 1h 37m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e06b3af38e0b4adaabf89e08f6a3b1c4 : SUCCESS in 43m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/82ad23c5e9594b9ba04fcdeb76f01444 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ae056fd3e374026aba114201cca2420 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23c4d18f68024d7cbdfed2abd79b09a5 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f975f72a20041809ceacaf2611dd152 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"29da379063ff52511bffd226c4fdeefebdd6fca8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-26 21:19:53.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/1/20250226205805/NEC-V-ISCSI : FAILURE in 8h 14m 41s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/1/20250226205805/NEC-V-FC : FAILURE in 9h 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0afaed81e20f38c107e95cee76aa896b3bf5c17f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:20:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe4da41b831d4ac8897f3dd7851b741b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f6913e62db54cf9a0bff458bd956d73 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37cf40039e7c4fa2a441475c55db096a : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7afe8d8f5bed4eeab3271309c41a3e52 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7fe71dc97614e2c8b70619a72b160b0 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ccdc06d60e34dab81b1cbdf54823e58 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/6027c013656e4c7ea6c367b36d55b213 : SUCCESS in 1h 00m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d4dfe51739b4ab995f7d9388c507e34 : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c94258bc4274a6eb8274295443ff10c : SUCCESS in 2h 02m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de257815818b4a84aaa84784f2c42bf5 : SUCCESS in 1h 24m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5fd99f775c24d40aa166713babdc8ef : SUCCESS in 17m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/825b01973bb240d9a6bab878bebbc363 : SUCCESS in 23m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6402090fccc14ae6bad758d2a96d4107 : SUCCESS in 7m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ea94cd252fd44beb43190c70e468f44 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2fac3b30964e473aba92681d7cb10569 : SUCCESS in 6m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/17561b5f6c3e487bad6e142c41d169ec : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7bf3454325e5457a814b03dff887b674 : FAILURE in 30m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11a36b7b0a80440daa55812735a51cca : SUCCESS in 8m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63e49f415f4c46889ba6ed25c1664fe0 : SUCCESS in 1h 47m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a6f56f4241e64bb88556b3a25591dc48 : FAILURE in 2h 22m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1d058c9982d4a2eb61ed440a25eb486 : SUCCESS in 2h 02m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9680834217e744ab9ddcadbc92048a91 : SUCCESS in 1h 55m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ef44e7febe4c403aa232fd4c5fc1f221 : SUCCESS in 28m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91d39406fe3e4a4aa7b6ca01cae37028 : SUCCESS in 1h 10m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/032de404bc5b4d98a14a8ccff94bcb1c : FAILURE in 1h 24m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ab03d1558a334a83b272a235c8c23244 : SUCCESS in 39m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d4895624874746fa966de0b132bc41b7 : SUCCESS in 1h 30m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fc9b0b5c71ea4a17b60078de5b22c747 : SUCCESS in 1h 43m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/faf353970c574b1ebfb195df9c34c688 : SUCCESS in 2h 59m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72ff8e77fea14490b66c93ac80b4b40b : SUCCESS in 1h 33m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c4418885cb143e496171c37e770d6af : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1f60e676bf437099dbba496951b6cda38ae98f78","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-26 21:21:57.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"c4e68518bc0f55aa190ca493acb3f8855047ec42","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-26 21:22:32.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"10311bfebdfd2bf809d754ed9774c152a4556960","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-26 21:22:46.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"107ce00c17d70761f3810f708c10c346a8819545","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-26 21:23:24.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"da75b1e49853f425bdb37d7cbf3b5914aeeae5c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-26 21:28:57.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/26_02_2025_21_23_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ac57ab3f354eeb1173a63818b74a9786a171c9fb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-27 13:26:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/3/20250227180623/NEC-V-FC : SUCCESS in 4h 08m 29s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/3/20250227180623/NEC-V-ISCSI : SUCCESS in 4h 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"7e5f49e1952c1093f290440186fc2322f37649d6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-02-28 14:34:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/3/20250228173832/NEC-ISCSI : SUCCESS in 5h 31m 35s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/3/20250228173832/NEC-FC : SUCCESS in 5h 56m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5bc22591065e8189d562c15c7ce0278b402c0148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-05 19:17:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e877321cd6d1d18eb035cc0fa24449f680dd046c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-05 19:18:03.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1394/","accounts_in_message":[],"_revision_number":4},{"id":"ce590b5ddcd6865a778d33fe499abe7bbb7d19ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-05 19:18:05.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"84e5ff5989563caaef5a6d75cda6dcbef160c263","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-05 19:18:07.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/nova for 941476,2","accounts_in_message":[],"_revision_number":4},{"id":"9b24a57243eb87f1bc97898c81fd085d46225307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:18:08.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 941476,2","accounts_in_message":[],"_revision_number":4},{"id":"bfcede8e11c28ccdd0cb07da6477d0fa393b2c86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:18:09.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 941476,2","accounts_in_message":[],"_revision_number":4},{"id":"30002be8d7e3425b6a9bc3cbb92533bfa2d21615","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-05 19:18:20.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/nova for 941476,2","accounts_in_message":[],"_revision_number":4},{"id":"d6e3cb0db95e6d715e4e9d69a3f6c0392414dcc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-05 19:18:49.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"5547c5f0832653f6bc69bb50df06fb915b47f0cb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-05 19:18:51.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"d1a6e1f417e994b46ad18b184e10826f89c321f7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-05 19:18:53.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"a22b1898bea3d07263d19b14e0a83403c94c806d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-05 21:29:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/424/.","accounts_in_message":[],"_revision_number":4},{"id":"4d2714608609eb52124e09edd85071ea773d0dc3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-05 21:34:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-4/FC/ : SUCCESS in 2h 16m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-4/iSCSI/ : SUCCESS in 2h 10m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"a0211b7c7d0ca5670b9463e5785dabb2f7cf64c6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-06 10:28:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/4/20250306151620/NEC-V-FC : SUCCESS in 3h 54m 47s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/4/20250306151620/NEC-V-ISCSI : SUCCESS in 4h 12m 00s","accounts_in_message":[],"_revision_number":4},{"id":"72be49ce45f7f483bb980c42c9dcb4070674cca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-06 14:47:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"10cfaafa3d4a357e2f0c358e9105d23bb93ac54a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-06 14:47:35.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1399/","accounts_in_message":[],"_revision_number":5},{"id":"a12b7ee76fb00300edbdfe2922e545ffc01c9af2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-06 14:47:37.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"4c15671f1268732698d1c01815485b4ffe7907fb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-06 14:49:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"d0221a76cf807ff2261a4f96444db26439103b8f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-06 15:03:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/4/20250306152757/NEC-FC : SUCCESS in 5h 53m 54s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/4/20250306152757/NEC-ISCSI : SUCCESS in 8h 36m 00s","accounts_in_message":[],"_revision_number":4},{"id":"675e8bc2325f95fba043d43e0f2ac5d09e544bee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:24:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/362cf5c256b74d05a6997bbd6510589f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b5173354c404ed09974014b78ac77dc : FAILURE in 25m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a68812bac41489e8efb98362c23ee76 : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"39b5700e8a4f86d84cbe06bf9550f2e5ab1d87b8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-06 15:57:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/942716 : FAILURE in 1h 07m 10s","accounts_in_message":[],"_revision_number":5},{"id":"63ae6a078c2d67754abf5eb5ee51bfab2b60d6fd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-06 16:04:42.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/429/.","accounts_in_message":[],"_revision_number":5},{"id":"2a53a0d7f73e23f22de2cb94b53ca315da1b8306","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-06 16:37:14.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a207e327ef2e4d44ac3bf7eeea353130\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a678bce97ea401090b6dd298c553f8e : POST_FAILURE in 39m 55s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/432c29b5a2aa47f98a563b898c342637 : POST_FAILURE in 36m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d6522b46ff04e8ca3cd9cf0e17c90ab : POST_FAILURE in 1h 49m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b70730c23f040a6a3fa5fc16fa83399 : POST_FAILURE in 44m 17s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7155aeedef2048ea836f98e1fd39e4d2 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3579ae8ca80841fd8f590657f58edee6 : NODE_FAILURE Node request 200-0000027929 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f2ae30ac4d6d495a8427b85e9fd98e10 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/881c8bdee2be4419ba8c2e460bb61aa7 : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job va... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"59e263811543bfaabffa3d81b867554530580692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 17:20:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29fdb525f9b0471a83ab8c7aacec60dc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3c853d3de414cf1a83b96c7d73e89e0 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac240d681154d2ebdeae63d3c62ec04 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca302b0ed35e42c7977f1bfa4f316a3b : FAILURE in 13m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9a57baabd9b4ff3a556e478bdf4ff56 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95fce09ca5145c1a9096180be8653ac : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/cea0ca2c35ab4a749f1e01f4ac10d12c : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24bb38dc54e740148dd30ed8f8957112 : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac009f58219842f88e7c0c4925a6a5e0 : SUCCESS in 1h 30m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0585fea2987044ab8b1534ff4af5d6ed : SUCCESS in 1h 03m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964b664bc7eb4fe4af3ac009a6d30e75 : SUCCESS in 18m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/433b59e67e464d45b0e34b78429ca3fc : FAILURE in 28m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/08bd41f9cddf44fb859c0295352247ec : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/509bf249b2db44b68a98cc4912e6eb39 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f505969b37c443fad16ab119acc6504 : SUCCESS in 6m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d2fee55dddd403bae9b825a57d4cc63 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/43a95c20117d4ebcb7d88fbd5458a57c : FAILURE in 48m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4277e6b1d65d4dc4b665b271f472bc4f : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/598e88e346ee4222a5cd10d0aadcf3ac : FAILURE in 1h 54m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dae0d543054b4f449e9683aad5753ae7 : SUCCESS in 2h 30m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1e3bec1d848540acbb6dc6b6201c56d2 : SUCCESS in 1h 49m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/399ccd0fb7434e449f62efc88b25a6f3 : SUCCESS in 2h 06m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/107ec16245424c2dba6ca09d219a3986 : SUCCESS in 30m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5bde93f37e364be4aaad52232ce74c81 : SUCCESS in 1h 00m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d88d29e888b14606b06ea6aa0533e4b7 : SUCCESS in 1h 35m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/066401eaa6ad4aa5b071a441f5d234dd : SUCCESS in 39m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bac69985e9c84d06a361896aaab62827 : SUCCESS in 1h 25m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9859002ccebe400fb1f9932277c07f07 : SUCCESS in 1h 36m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9229ba6945af402e9f7b17ce3f8218ef : SUCCESS in 1h 33m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cf052331aa4cfa9cbe6d5361637147 : FAILURE in 1h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef5cdb9a09ef4165aecb59d605407e73 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":5},{"id":"3965a10afa388255c19d209b965960907dbea84c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-06 17:47:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/29b04ae1023244349e61ee6b24c009a9 : SUCCESS in 1h 07m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/78179c505d76444c8e74afb100cc7d0f : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":5},{"id":"2b7a099dd8b4e5633460e7dbb9a048c3faf93531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-06 20:57:05.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5176bf40142693acf24ff04c8e809e5ae2f6d70f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-06 20:57:16.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1404/","accounts_in_message":[],"_revision_number":6},{"id":"5f1399300c455e9e2285e11439d4521c2a71df92","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-06 20:57:18.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/942716/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"2deefcb4fc83d3ec53ed4de2b1fa090686609bbd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-06 20:58:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"66e3a45286d2726b42cb1701c0246575e3805283","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-06 21:14:05.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-5/FC/ : UNSTABLE in 2m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-5/iSCSI/ : FAILURE in 6h 26m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"6ca738e3bf8f11077278d4dcda5c6cedc80a4a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 21:15:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07acb2d5209345e890367fc48c5443fa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d29ee94d60844cfbd9bf430cd0c5360 : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07b8b866049f4fe1beae4ec424ef35f1 : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"eed7cac2c9ee757254d368c13bd8c25a667c1943","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-06 23:00:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/254a634e15cc4a8e82c4cf4c77270a43\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/787f04b4d8974b87b2f1646f4c285ec1 : SUCCESS in 42m 54s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ac2cdeafb2e44d88b462cebb547ccdf : SUCCESS in 39m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/78dc69ee79ed4ff5a7cb6dbff6f20a43 : FAILURE in 2h 02m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b927135a54c4ebf84754ab38a81960f : SUCCESS in 44m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/09cda9abc012488ea37e304a0d4f0c3c : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65e67bfd70ca44b489886ef8edf720e6 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43ad26ee470f4d44a94684bbf9c829c4 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/acb4338de8894420a30472e53febe960 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"43b117db401beb9c0aaf90a90ec0465d6ef4888a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 23:44:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb784a7829174720b36df344553742f3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7a7a34f6b22b4729bb6e4b78923935f8 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253f0d01bae543a59e7fef20a9ba1586 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/865c1b369bf648e491a9f2bea4813aa4 : SUCCESS in 11m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb09be78abde473291eba28765499c15 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38525c6558b24475bf7c91db51d168fe : SUCCESS in 11m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f0fd5e56f45d488b845a59f6d8ec700d : SUCCESS in 59m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/09755a05b2d64fca8877f7b39b493c16 : SUCCESS in 46m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/436c236ae7634a738411c6f7f649af69 : SUCCESS in 1h 25m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2afaa0fd2b594790b6ebcf102a6ab05e : SUCCESS in 1h 02m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecd7f93643048afb0eb0cb79f1cb302 : SUCCESS in 14m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab6ca079f02c4aa7b6d8400ea3356dad : SUCCESS in 25m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/016ea3af86ef4c6699098698a32f3c9e : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec51cec558e74810bb89125616755ead : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5ac9bca1197c4afa976e3b6bd00022b5 : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e22267eac8864830a6c4f3f42ba5ac18 : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce00a704fda94cecb01ff17704b95712 : FAILURE in 48m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa3a20dcd5cd46c0ae9925485a813424 : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a7ea7de50ed4ebb9fa8b4925ed25057 : FAILURE in 1h 47m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/92821d837d6440789895b861e5c2f002 : SUCCESS in 2h 30m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59be08abe92f4a1eb6943cc9e5d619db : SUCCESS in 1h 37m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0da46092eef240b9b75c9e3454e9fb3d : SUCCESS in 1h 36m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6a9f667c98e643dabcfeb0c871199c43 : SUCCESS in 26m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5f15f7a4d7d4f97a1c13f3f98efe306 : SUCCESS in 1h 07m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/52b0411d480045c2b34d237973283197 : SUCCESS in 1h 23m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ffd0e1bd3cf42198dd4a91aa29b5acf : SUCCESS in 33m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3fca8e8dfac4f978174a026adff6fe0 : POST_FAILURE in 2h 40m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7f024fdf841f447a91eca31191783f4f : POST_FAILURE in 1h 56m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cc1bec7eff14b94a50d65793c4c82c1 : SUCCESS in 1h 36m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61b2c69c976a4e89b9f772a822c7860d : SUCCESS in 1h 51m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb28e55e6f104c2fade70c04288c74e4 : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":6},{"id":"4b787b902fcb02ee7dcc339e07b0f722ebdabc46","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-06 23:56:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/942716 : FAILURE in 1h 06m 50s","accounts_in_message":[],"_revision_number":6},{"id":"3aa4205ea079c7d6c64162317a4d91de4e0974d0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-07 00:01:47.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-6/FC/ : UNSTABLE in 2m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-16-942716-6/iSCSI/ : FAILURE in 3h 4m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"6371c71ee736bdad440e87eaeca3bde34858783d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-07 00:17:46.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/434/.","accounts_in_message":[],"_revision_number":6},{"id":"033845c1c21587df57b0140fe3791af99db3b0f8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-07 01:04:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/766988d1831d4bedad4635166b5b6744 : SUCCESS in 1h 14m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3525d4c919924010b81e30dd9e8ba4b0 : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":6},{"id":"9241a86bc23b9e72c1e7cb30e8c90ac06b78ead5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-07 01:33:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e75b011a4bbff695147c01c203dca0f40973fd5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 01:59:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/496f8fc7188d4f73949fe3045e5a1bfe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cb7d89855d846039f27d3b45a746b57 : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c76653641e9440448324b90f4d3692f8 : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"58e949e68356566772cc564ae3eb3074ef4f634a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-07 03:43:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9fa0814debad4497b6f8f61dc409a607 : SUCCESS in 1h 15m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/db12c6a2fa0c41b1981803460d90c9f5 : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":6},{"id":"4660d2766ff69032b9d841fc346060c13fde6d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 04:23:26.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6555529cc1214abca6eb3255a6869501\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44e26efee4534f4c8b368b43304afa09 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad8127f8ba34805a54e879032c0a610 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee6b4059af9e4241bfcfb04b53cc0295 : FAILURE in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73d2d7656c2e4ecca80a35e6ac1436d1 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f25ac39b56b45b4bcfc9b310c91fa62 : SUCCESS in 5m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/78cc5722aaff45ea9e482522fe72d50f : SUCCESS in 29m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a581c780fb0480fa880fc4c5b343655 : SUCCESS in 26m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ad348ec1d81408b8f3df1771b538bc6 : SUCCESS in 1h 27m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba1ba16ee174dd78049425cc21ae54a : SUCCESS in 1h 16m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2032dce8133e40afa6623110869fd945 : SUCCESS in 12m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9734ad85c704c63bd24f78e8032064c : SUCCESS in 17m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf7ed426fedc464fbcce97d40cec89e1 : SUCCESS in 3m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e56b073d61eb40459fbe21a56ebe5992 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9ec8cd0ff0f4a5a99282112b657b2b1 : SUCCESS in 3m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9173611fd1fb46c5878ebe851b083002 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51210c6b80c549d3bf64e0aca249bfa2 : FAILURE in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80755da6d460435d90f20a130136555d : SUCCESS in 6m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e9e9d866bce4dc89cc1ee662dcbc976 : SUCCESS in 1h 43m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff355aae133e403a8e26fe982f79c34e : FAILURE in 2h 43m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8833c86d35a5419f9ec2e198c880ac67 : SUCCESS in 2h 01m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/80770e5ad3ea4ac992bbbf0d6309672b : SUCCESS in 1h 32m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/66a1a660b87f4ad380bcc31a59e5cc55 : SUCCESS in 25m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/622c3b8fe5574fc486104ab80780a32c : SUCCESS in 53m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1e1e490f593421a834cb23d7361d51f : SUCCESS in 1h 40m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/59b558d5d9174c5082ba5804fc69311d : SUCCESS in 48m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0b65838c4be4c7db30c1e94449f906c : SUCCESS in 1h 21m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c20b475653764a59b2c0a8a09c9f33f8 : SUCCESS in 1h 31m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c116546ff60b44aa9d34f405dce73d53 : SUCCESS in 1h 59m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d82093f51e64309b69f1f1b55d79ddf : SUCCESS in 1h 31m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43338ec65f6844f4ba86d8b4bdd0d27c : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":6},{"id":"3638e979f74b5553be36361868d86cb51ff05287","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-07 05:31:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd75bfa486c39202de48dca186f14bceeff5a603","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 05:53:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21322525dda24b04b93c734f2722be32\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d3155aaf6b3487cb919c400a40cd31e : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22a55eba99304aad987c797f1f521a40 : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d8bc0982fd9197a2d0144ed2826530821d30cfb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 08:06:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad8b9d2a93d498baf883fc18547cf6c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f73a1cea539e439eb9149ab22fa2d878 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6c4c64743a450caa4f8cc41111c302 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec0261449fda45728e63bf6898bebf71 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/530b0c99bd864ed2ae5280ad1e9cee8e : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3633e87ff95e4eb5affd1c7e2b1d29d1 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0523c795b72741379d3f5b5b3d721342 : SUCCESS in 53m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b62ae054d7a4ee885af1ce09210abb4 : SUCCESS in 52m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ac574e3a0044876bba293b1904dd45d : SUCCESS in 1h 20m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e5f5674f4da4fc99ba41e271c6f8b0d : SUCCESS in 56m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf21c78fda5245069cee0737dc41ee47 : SUCCESS in 12m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f972d0865fd844b98a084f0b360a42ad : SUCCESS in 22m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/82fbb39b46444d2597011867bab535f3 : SUCCESS in 2m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5354accbf2441adb9cacbb241efb5f0 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f97c39d7f934a999d8bf8c6d35a3539 : SUCCESS in 3m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2b21436c97d74044b004666d4c44b2ff : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f13e56e36cf4503a8de548ec85c83b9 : FAILURE in 19m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db30b934a9f744d1899dd52b99d3fdde : SUCCESS in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d9883652022472ca768d49c9d1b91e7 : SUCCESS in 1h 44m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2589d03926e84901b4576f9543a12206 : FAILURE in 2h 29m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dba33d10c58c45ea819e487e43b1179f : SUCCESS in 2h 02m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7f76a0a2a174886b867425f7847266a : SUCCESS in 1h 46m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f47cd1547e9d428294812cb1c16997a2 : SUCCESS in 18m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8f39a24a98d4c1995052d94c142e8fe : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a736edb8179b4163835354852dd9bf58 : SUCCESS in 1h 46m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5483c1bde024f9cb0744f6550c8c046 : SUCCESS in 57m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/acc52558d25b41e2be2d4e3eadcb9680 : FAILURE in 1h 01m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ca74674f5bc3414ead074af6d576f9b3 : SUCCESS in 2h 10m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b38590c41f0743aea5920663e0b7ddf2 : SUCCESS in 1h 09m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40bb8f34012a4466880bb8c317ec67f6 : SUCCESS in 1h 23m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63a718b12ae74804a7fd89859a632339 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":6},{"id":"614a8467875893276f4aa3a29426593cda5bd464","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-07 08:55:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8421588139f44fb7a03f4b3cd6bf38ce : SUCCESS in 1h 19m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bf1fb634c9fa44539751d896ae20c0d0 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":6},{"id":"b026e27d4fbc9d3485eca84a52c1608b298fce49","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-07 18:17:37.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/6/20250307172810/NEC-V-ISCSI : FAILURE in 8h 13m 04s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/6/20250307172810/NEC-V-FC : FAILURE in 9h 49m 24s","accounts_in_message":[],"_revision_number":6},{"id":"810d1859827f142d555662a8ef816bac80f42c5e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-07 21:21:49.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/6/20250307204820/NEC-ISCSI : FAILURE in 9h 10m 36s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/16/942716/6/20250307204820/NEC-FC : FAILURE in 9h 33m 26s","accounts_in_message":[],"_revision_number":6},{"id":"5db4d93af95e1432f9ab5ecb3f9d0c76acb5c215","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-08 06:48:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3a954dd52eabe05a70f2e96fb96fbee2cae35618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:11:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfc2d63a27e4d6c9455bf354d01c61e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/397d645fa0154779a80a1766faa9c519 : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc368d6caf7248c0b675a1e38a58eb02 : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"592677c1c48db41143ba7c1e0fc6aefeb0a137d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-08 08:44:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4e9483ca3c94c36ac7aeb2f419b32c0 : SUCCESS in 1h 15m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9501fcf9975d45919a8e22e83bd37e8e : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":6},{"id":"80c9619d7eeb9fc2012aa1d721b5e47795888d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 09:01:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb7928043154f9281cfd134e2608a00\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a025e8ca70134d759567101f572cf3be : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64e3c31334f44baba3b07d8024ce7ee : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33bbbf6016d44aa8e604a886da4b257 : SUCCESS in 7m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfb28866bda74f838a559f1c90d84b51 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a6598cead641fca005bb97b6c73ed2 : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/cbcf465cbcc44012ae7431191cfb1697 : SUCCESS in 46m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b804f577ac5846e7b59ccf9ec89cb0dc : SUCCESS in 44m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cd7867ce61240ea8852fd86186a7d96 : SUCCESS in 1h 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3307ab404684431afabbe9253458720 : SUCCESS in 46m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e42dd4c583487699e7a5c0f158a4b3 : SUCCESS in 12m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d04e32dc4344f1abacfe29b957928db : FAILURE in 21m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e69146f5a2f4c5682c875ee96e26994 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be59d8186762405b96fe221c667389c6 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30449e2736cd40d783430a0c04ba08bc : SUCCESS in 6m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ed217e057aa4c6a902174d997d0afae : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a75c814d1a0646bc99847b5ec11fc49d : FAILURE in 26m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da7ee1d167a14c42b0a4799781257dae : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c081c1080ba4b7e9ddc5a67e5eb6c6f : SUCCESS in 1h 36m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/538796745f82443d8ed9f27bba13dab9 : SUCCESS in 2h 11m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c15ad8fc76324882860bdaecf890c7e0 : SUCCESS in 1h 29m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c6235cd1e64d461b8f22f9fa603581e8 : SUCCESS in 1h 29m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fab2e6a45ffb4d30b51fa6930d7ad643 : SUCCESS in 22m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79a91d76736a46a4b9565524b27c75e5 : SUCCESS in 49m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d584bcd270c1448f9ab73aaaa119087f : SUCCESS in 1h 30m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52e200fa93bc42d4904eef0200df554c : SUCCESS in 29m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/17130a41f7b741d98caad6bc603d84b0 : SUCCESS in 47m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9ab496a4a5004d65ae947e1a8f1113bc : SUCCESS in 1h 00m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c02f7998981d4057a677f17f3844f245 : SUCCESS in 54m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/096ed75c75a5418bb220053931128b63 : SUCCESS in 1h 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/869c7ceda24b4bac86c6a9617d16fb58 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":6},{"id":"13fbdbf37499702dfaf40087b86b25e49084e56b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-03-08 09:02:12.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"5e1564c38693c44142d5d1596a71d0b23bfbcc38","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-03-08 09:02:22.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"b18798517c5ad1b3c6c2ef000b7e727879e216a9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-03-08 09:02:57.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"d0d9db45e3c6afc84057d916efc4582c29fc70e5","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-03-08 09:03:08.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"9a14c6e853a1082730f48b9ca2f4f13fd5f6b54c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-03-08 09:06:42.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/08_03_2025_09_01_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"87058af018a9beb41a6b3eaf73ba1eb3646e7e5e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-08 12:02:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c8e9cfa6e59d49e89cc0b4b4d9f89f6b5ca3eb8e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-11 23:39:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/942716 : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":6},{"id":"13ed3dc8af883e1da55cbeffd02adece6b2f4223","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-26 11:31:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ad2c6478452e2f1d9c9a2bdc174ed6ef9d06a0b2","revisions":{"5b4ea52117a2023be0ba948cf642a91b17716423":{"kind":"REWORK","_number":1,"created":"2025-02-25 17:28:06.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/1"}}},"commit":{"parents":[{"commit":"3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c","subject":"Merge \"lightbits: add qos support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-25 15:08:30.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-25 17:11:23.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b4ea52117a2023be0ba948cf642a91b17716423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b4ea52117a2023be0ba948cf642a91b17716423"}]},"branch":"refs/heads/master"},"6e129d424f9cba4ff634092b39f3870e9146606a":{"kind":"REWORK","_number":2,"created":"2025-02-26 14:19:04.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/2"}}},"commit":{"parents":[{"commit":"3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c","subject":"Merge \"lightbits: add qos support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-26 14:18:54.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-26 14:18:54.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e129d424f9cba4ff634092b39f3870e9146606a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e129d424f9cba4ff634092b39f3870e9146606a"}]},"branch":"refs/heads/master"},"b9af72deb3097dd2cbab6e5a0c9f3138a0167703":{"kind":"REWORK","_number":3,"created":"2025-02-26 18:14:34.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/3"}}},"commit":{"parents":[{"commit":"3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c","subject":"Merge \"lightbits: add qos support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-26 14:18:54.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-02-26 18:14:20.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9af72deb3097dd2cbab6e5a0c9f3138a0167703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9af72deb3097dd2cbab6e5a0c9f3138a0167703"}]},"branch":"refs/heads/master"},"dde2a827aec62aae6e696ac5a9260b92c26becbf":{"kind":"REWORK","_number":4,"created":"2025-03-05 19:17:51.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/4"}}},"commit":{"parents":[{"commit":"3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c","subject":"Merge \"lightbits: add qos support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-05 19:17:12.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-05 19:17:12.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dde2a827aec62aae6e696ac5a9260b92c26becbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dde2a827aec62aae6e696ac5a9260b92c26becbf"}]},"branch":"refs/heads/master"},"7405ff659ea9aa303856062ba2d1505e1541bd60":{"kind":"REWORK","_number":5,"created":"2025-03-06 14:47:25.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/5"}}},"commit":{"parents":[{"commit":"3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c","subject":"Merge \"lightbits: add qos support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e7e62dddd7a8ebb53a25f3ef137418dc927ee2c"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-06 14:47:07.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-06 14:47:07.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7405ff659ea9aa303856062ba2d1505e1541bd60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7405ff659ea9aa303856062ba2d1505e1541bd60"}]},"branch":"refs/heads/master"},"ad2c6478452e2f1d9c9a2bdc174ed6ef9d06a0b2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-03-06 20:57:05.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/16/942716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/942716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/942716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/942716/6"}}},"commit":{"parents":[{"commit":"f2b4970cefa585dde1937ccdca4f2d44c6725b9e","subject":"[NetApp] Fix latent issues in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2b4970cefa585dde1937ccdca4f2d44c6725b9e"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-06 14:47:07.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-03-06 20:56:41.000000000","tz":-480},"subject":"Show migration_progress in volume details","message":"Show migration_progress in volume details\n\nConditionally add a \u0027migration_progress\u0027 field to the volume\ndetails API response whenever the volume is migrating. The value\nis a string that represents a percent completion (\u002742\u0027 for 42%\ncomplete).\n\nThe conditions for when the \u0027migration_progress\u0027 will appear\nin the response are:\n- The request context is an admin.\n- The request specifies the new microversion (3.72).\n- The volume is migrating (migration_status is \u0027migrating\u0027).\n- The migration progress can be determined.\n\nThe migration progress can be determined when:\n- The volume is attached and nova is copying the data. The\n  migration progress is obtained by fetching nova\u0027s attachment\n  \u0027swap_progress\u0027.\n- The volume is using the generic migration path (not driver\n  assisted migration).\n\nTracking migration progress when a driver is migrating the data\nis currently not supported, primarily because it would be\ndriver dependent.\n\nImplements: blueprint migration-progress-in-volume-details\nDepends-On: I3970e3ebd79d729bf17456fe6cafb8f9abaef5ab\nChange-Id: Ib574ee5a6ca2d4da06140ff9f2d7abe36e8afade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad2c6478452e2f1d9c9a2bdc174ed6ef9d06a0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad2c6478452e2f1d9c9a2bdc174ed6ef9d06a0b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
