)]}'
{"id":"openstack%2Fcinder~945614","triplet_id":"openstack%2Fcinder~stable%2F2025.1~Ifdfcd480b72727ec182d5a6954c706f365247edc","project":"openstack/cinder","branch":"stable/2025.1","topic":"cgroups-v2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-03-28 17:59:00.000000000","reason":"Change was submitted"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-03-26 14:49:06.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ifdfcd480b72727ec182d5a6954c706f365247edc","subject":"Implements cgroupsv2","status":"MERGED","created":"2025-03-26 14:48:41.000000000","updated":"2025-03-28 18:00:48.000000000","submitted":"2025-03-28 17:59:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945614-cgroups-v2-stable/2025.1","cherry_pick_of_change":938490,"cherry_pick_of_patch_set":7,"meta_rev_id":"44ece534530abc89ad8822c1ddab94c8bcfe2cc8","_number":945614,"virtual_id_number":945614,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 17:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-03-27 01:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-03-28 12:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-03-28 12:53:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 14:48:41.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"CC"},{"updated":"2025-03-26 14:49:03.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-26 14:50:22.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-26 15:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 16:52:25.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-26 16:53:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 17:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 21:52:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 21:53:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-03-27 01:25:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-03-27 21:41:59.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-28 12:53:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-03-28 12:53:50.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"7a40aab6ed95db671c541045439e40f080df0d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-26 14:48:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5b7d39c5287fcadda8804f60ade93773e5d33d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-26 14:49:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"41facf689f03d5033c4f477e8cd251a7c0cd969b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:49:03.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1585/","accounts_in_message":[],"_revision_number":1},{"id":"b8bacc48e0775a9e90e978613181f9e7ac1f5bba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:49:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945614/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945614/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"f68193b59efc506268675f297ebaa8a5a29d6227","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:49:11.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1586/","accounts_in_message":[],"_revision_number":2},{"id":"f82828da70cffb7dac1202a7db475f31961330d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:49:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945614/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945614/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"fd2c7dbdcb98c079c01276b76d41e94a986cb992","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 14:50:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c4fbb03362ebd4c4d6ecbd7b25e2920fc38a9023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 15:10:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d039f47dbc64a8e847a1fc066ccf9f5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb402e71983c40bc8ca0ac60e76153bc : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/943040bad91c4675943e44c8afa52337 : SUCCESS in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ec076972c295e77e4536947b385a791af8d827c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 16:52:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945614 : FAILURE in 39m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c444e8ee03828ccb7446f35603ba96525e04dfab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 16:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945614 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":2},{"id":"029fd5d76e67f44fe35a71ceb46e01a5ef5ef867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 17:24:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1d30b89fbbe456691bf0543cd530921\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082a659523de483c8353f943ea7cf686 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9362799568394561beaa98f797494a58 : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/638f18d6585e40a197abc6d2c43e47e1 : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46670b83d042493b8de2401b437cf689 : SUCCESS in 14m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/89b2ac05144546308c67461b8f671fc8 : SUCCESS in 53m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/202b573258cb40b69a51581b85f2035c : SUCCESS in 1h 21m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebf897a7eadd4b36b62ecd725889cf38 : SUCCESS in 56m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664a5c5b56dc43d9aff8dae439ca08cb : SUCCESS in 18m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5846539f4e5f40d79bd07f1d041f37a1 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9bb737e68e624672b59704733eabba19 : SUCCESS in 7m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f314a4d4b7e74fb6b2e01a2013c0ba69 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bc3ff7c3caa4c32a7d83f61fd543dbb : SUCCESS in 8m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7cfc840183b5421a8db60a05c6dae8c1 : TIMED_OUT in 30m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1930537945504f8486af45c05fe380e6 : FAILURE in 29m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bad4ebb0286b4c9eb09e9a368a5280ef : SUCCESS in 8m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/50c7eaefbce24504b6cf6bc2bf2e7053 : SUCCESS in 1h 58m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/12653b9a2e7a40a49d9f094434760055 : SUCCESS in 2h 27m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5efabe21aba640b6a996195f83352bb6 : SUCCESS in 1h 55m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0972f30890a646ad9bfb22a215eed117 : SUCCESS in 1h 37m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/298dc9a6bd914a4893cf4b68689e82e2 : SUCCESS in 32m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8fe61782c584949ba9face9d889227f : SUCCESS in 59m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3bf87312432043de9395e07a4ceed302 : SUCCESS in 1h 31m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/956060958d6946609fd22d36e5397911 : SUCCESS in 37m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0fa2a876e5d449ff9e6938d58de920d3 : SUCCESS in 53m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3c96acee582f4200b71401d6d5c6b24a : SUCCESS in 1h 26m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e96cbe9cd8ab486e91ade264093296c4 : SUCCESS in 1h 19m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abc786e7a05347b99c51af9d2cb8dba6 : TIMED_OUT in 2h 02m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8e031c5713745e2a24e5305a27b8806 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9bc4431b1d26dca9345bf4b29d9c24f3a81fcd13","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 21:52:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fe880e9db42f52a18f3ea1392fbe1656a97ddcb1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-26 21:53:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79cd9a42d3e543ec3d6e347b2ed3f8469f57bf79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 22:14:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e82f0fe90d24ccc8b6cc6cce16b0370\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b08a22aeae14858b57a9e60421302de : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4eda4c6b6054416b9637a85a4e083c7a : SUCCESS in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"598d0b9f1a13df44985bc86b33ec1c9d16f771cc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 23:09:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8dfd0fe5fd79a610893e5108f65e62e6066ebbb6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 01:25:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc226da7bb3c7c21283f75c95fa956108783bfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 01:32:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e67f7109e984dafbe88ff50d94cb22e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5115249f42844b7a89688780c8b66d72 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e798ec1ccb14856a2b3a7e40a1c7d86 : SUCCESS in 12m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af477e141e354edf939177f3f0ef1cc8 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16f6646d75c4ae99518995f2ef00cee : SUCCESS in 6m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/78cb36f0d7df48928a398a5c034dc6e3 : SUCCESS in 47m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72ef06bd6f8646b5922493ee22e4a1af : SUCCESS in 1h 35m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/987a9b4fec6f42b684219165b7a2a6ce : SUCCESS in 53m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b70e22432eb4583898c5a66e895404b : SUCCESS in 13m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad2ceea38ab0429fbe8be77e5b62cf3b : SUCCESS in 25m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b34bf7f0e6844d9e8ee8f8ec62b12287 : SUCCESS in 6m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78d653e8e31945debe3bb6418dcf88a0 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7fa05167b6d469d81a720e2276cb48c : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f0ca6d744ab4d928a5cbe165c164bf4 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf576ec6f9954c4eb805494c1bc842db : FAILURE in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b10c05cdba88466db5906af97c8e596d : SUCCESS in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/375c0413835145f6be58ca1c4ff94147 : FAILURE in 1h 19m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d61553afd78418797f3cf14e11c131f : FAILURE in 2h 07m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28498a61a08f418eb6c7df4a0d65af1c : SUCCESS in 1h 43m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/da141e983b564f078635313a8b1b86be : SUCCESS in 1h 50m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9feed264dff344439374f22fa19c18c8 : SUCCESS in 21m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d2444026f0d4891a796a3f22b043794 : SUCCESS in 1h 06m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ab108c80bfd4c65ae86e23fff680f3a : SUCCESS in 1h 40m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eea7bcc979454c789af5989fb58109d0 : SUCCESS in 45m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/739869d6c2234f738594b437704f580d : TIMED_OUT in 3h 33m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4fb461dabf4a4ff1991ecb10afd9c27c : SUCCESS in 1h 45m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49ae727679f847df8028f6c8bb3b96a8 : SUCCESS in 1h 34m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af05645d42ae4d848f8d4be298be7d44 : SUCCESS in 1h 23m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd85e5c2f1041869c09ee54cc07427a : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4e4ff2f61303ef0c6def773a993a848309e28315","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 02:48:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fdef66e762a789929e52615bd2b7300bb843a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 03:13:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f4f68ef4a34a9287f0cc8c2f47495d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/007cdd577a65495f8889389df791bdc2 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a8cc87b1dc1476f950368183b5e1938 : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd6e9f1eb10d00a98a8e1739f6d37519fd2447be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 06:19:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da505ccaba444051a4d01bf22e3fa7d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749c9a105f724e36a040c4876fde4299 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dac5c3a32bf4cb496aa26ec89781d58 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e6d4fa450fe4e75ab346e27bda1c7f8 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8152abc845e47dea6e666c3415cc462 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/fa6e70f6a2e74eb9b42f3054e6c8881b : SUCCESS in 53m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/051397ca95e5497a91dc2825862f923d : SUCCESS in 53m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c00b4a3f9814d60b14e3c53ab2719f4 : SUCCESS in 1h 25m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24569a5a7c614b95bedfb628e552a193 : SUCCESS in 56m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6b7b65b77f84da38c249d85ff225e5b : SUCCESS in 17m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8bfe8f127d84e3d8955aa6960c5e545 : SUCCESS in 19m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b23ea87ba8d748d9bc11ce4b177eebf4 : SUCCESS in 8m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/186810990d2e49eb88bcd2b1d41fbe32 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59de2cfcbdc0479c931853ba5056ffaa : SUCCESS in 5m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/85994ea2234542f6bfb353cbddeb809e : SUCCESS in 5m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d9cedb137454ec59a6f77fc1c86d9f5 : FAILURE in 31m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a54c044331c43b794a2cb91a0c7db1c : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3998ab26f4b46fd944017814753575c : FAILURE in 1h 51m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/990bea566c964e2ea7d0d0a41070597e : TIMED_OUT in 3h 12m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/44df196140634663b2a8fb94b0ec6612 : SUCCESS in 1h 46m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4a2f0cfc39e9497cb235f0c3caf19a2d : SUCCESS in 1h 43m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/23c00f1be60d4141b11c43a0bf19db85 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf47729f23e245c5a77b178720c63984 : SUCCESS in 31m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80b7e719d4cf42dda7a1d18fa3ab2bce : SUCCESS in 1h 42m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d16a3a7cea44c20990f979224f04a09 : SUCCESS in 35m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/099816e0d5c9488e8cabb298636f092a : SUCCESS in 1h 29m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a263cb24a4cc4e4c81dfc36a8ae2c82b : SUCCESS in 1h 34m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cac533df9094322a8debfc412e07286 : SUCCESS in 1h 23m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3e42e020125440798f2a0bdca2972c0 : SUCCESS in 1h 34m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f040acf29614173a9e089f99d846945 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d1ceedf59dbc1bba707eb9711035dab5782767c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 11:11:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d904c82e47498d6f3533975485058267379c6612","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:32:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dca895f75e94b2093c05d45fcace608\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84b20f4b5b3240c59c9bb4b200417fab : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c22a7025fe18443baf37a4abe57c9f2a : SUCCESS in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec84e7306bb8dadb6bd2846fb008333aa0170514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:50:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d70c40046e0942b49cfde94ea3a2bd32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5f4b0ff39a40639a5d30898c620940 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1514d889c4437c8d09100d9a640f68 : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0157310931934e6c8847555e400a322a : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae73d50c11fa4d3cafaaf2f6e6f117ef : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/dbc73dfc607740b5808a2e0f4d9175e6 : SUCCESS in 59m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0a9f559d9ac4636a739fcf8c28b8a8c : SUCCESS in 50m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90a1b21f5ff34647a0ac364618a4ea48 : SUCCESS in 1h 32m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/034f79062a854fca89fab87c77739a8e : SUCCESS in 53m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06dcc53a53054ca1b75ffab541947d20 : SUCCESS in 18m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e9e64e3a1f842b3bbf62b199f8246be : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c3b83d6e29b42e88b725f61f2c49265 : SUCCESS in 3m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/103a4616f30247958b755bbba8806e22 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/050942259772474bb97a3c212a61570a : SUCCESS in 6m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f884f56ec974d5f91c9505196890d34 : SUCCESS in 8m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7eaf12dc309f4c13a4585ba403c8caa2 : FAILURE in 30m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03ba85b9be1c4cbfa5130a4ca53dcf23 : SUCCESS in 8m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd3e14a5132241f99340b6873c46519d : FAILURE in 1h 56m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/878d252b5d6846428697aa7b83620f94 : FAILURE in 2h 22m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dd69a9ae0ff483784ae5a4922da83df : SUCCESS in 1h 54m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cf1dba5ecc0f4303af90beb21d619668 : SUCCESS in 1h 58m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ace80208bc9c4ae5b0316058c6c7d1bb : SUCCESS in 34m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29ffd092cad4473d9367d8cc436c3981 : SUCCESS in 1h 07m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48a53068432c492c89d19479b6ab9c4b : SUCCESS in 1h 32m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04b9befc2cbc46a389881837175a643a : SUCCESS in 43m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d92622634d549b19ee2e2bd62693fc8 : SUCCESS in 48m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ea7ce8b4fd194e1b8b368052b42cf9e5 : POST_FAILURE in 2h 30m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8851d799e4649d8bf3e32a3cbc2a8fb : SUCCESS in 1h 47m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6ea68954a2c4990b77262292918e12c : SUCCESS in 1h 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05d496dae4f747b996528ee683813098 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e90b60a067a48887d90953e69e940a349237f187","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 14:01:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed100fa2e097322dd46c85356190c25e8a926fc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:21:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f06502b15140858043c1af5fa8e328\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da7cf4b1a80e4790972cfb7573790f69 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e75e2a4a75a54f6c96c2491c871675cc : SUCCESS in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c3171ca6d12e92ac03c698fd7708cb827ef7d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:36:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af3b83703455465a998b5a629d440881\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590b869974c643d8aabb080df12460c4 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff98c8295134034aaa3608d36f49ee5 : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac87b4ef5253478a901f2ee67880e1c7 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00dcec610e3471d99ab87a2edbddd05 : SUCCESS in 17m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3b2d4e4d0d4f45ff8a74bef36a2a25a0 : SUCCESS in 25m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b744089c9ec74825bc6422180c11f5ca : SUCCESS in 58m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8bd6e9ca7784b1fa1603255c635ce9d : SUCCESS in 1h 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0af99107be1b495292a122a8cfc59bc2 : SUCCESS in 58m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d732e09d0e3f4b5fb6f407240e0b6419 : SUCCESS in 19m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8d790bfc35a43d1ae2fd230069cc859 : SUCCESS in 27m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0321929ff9cd4c61ab1b53b8f857e2b9 : SUCCESS in 6m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ed84e20a54e47d983aa7e954f949dcc : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c9a1065ec8e467ca35d3c154069d349 : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6fd54dcc61c74475a817ecb8e146d6e4 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3fd55908e4d49f3aa231982b14572e7 : FAILURE in 47m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27411c41b2c64eef92334fa77253c84e : SUCCESS in 3m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86bb8ced23b746cc8bae8f82d2a7a68a : FAILURE in 1h 10m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aad88f22fe6349ea9cac185dae95562b : SUCCESS in 2h 34m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e08500b5e1224a11a8e940618d0be8a2 : SUCCESS in 2h 34m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4e884633a4794d1bab054d424d8e80b7 : SUCCESS in 1h 52m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/332116bf5c4146b398388f0d8cf38c62 : SUCCESS in 16m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/679c1ced0dfd4145a44c5955c6e48769 : SUCCESS in 1h 02m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6911ef03fd5b475285aad3615562ec57 : SUCCESS in 1h 36m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef6f23371ccc4209ad080fa4c62b8b99 : SUCCESS in 48m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5d91025edac49af957b24e8084f9c18 : SUCCESS in 1h 27m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a58a1a7b9ec1411cb830914b0e15ce94 : SUCCESS in 1h 35m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df5f235615504d5995c796c8be539a40 : SUCCESS in 53m 21s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29ed61b16c344973bcc20d07c5494642 : SUCCESS in 1h 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90bcbd1c5e034dc9b4daed2b2e056fb8 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a95723b4a2f366c7b6a5d50abeb34ab71329f247","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 18:02:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8f1bb5e7f7c05b17ab1b2d86729bd7f3e1505d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 18:30:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a630b0ab40424a854748ccbe50d079\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5cf1f6177b7745c394b775e9c357ece2 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59057b83edc749de8730493da7b5039d : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"605784533d794cb24b5865a6c74e04599a3f63db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 21:25:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ef5aced35c419ba1cafd57fc5fc23f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbad58af11d446aaea7c76ac52c4a31 : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5587c4fa9ad844ad8f5553b040bb8379 : SUCCESS in 12m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/680864674e00480eaf878c6ed6ca4677 : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b3e24cd0d91498da7507338ecf69c2f : SUCCESS in 15m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/cc547e578ce14585b9c32f0064cdc0d4 : SUCCESS in 57m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99da4efe92794d7ba3f1cb43c62928c1 : SUCCESS in 1h 06m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fed47aa77994f9fb4c5c2e6f0c013dd : SUCCESS in 1h 59m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d414c3c02d4e4f209613f8d3e575cc99 : SUCCESS in 1h 03m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945cbb8efc534cebbc39ced67962a749 : SUCCESS in 12m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0d4c79ed461e4e01851a97b665c81869 : SUCCESS in 27m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/679381e130e74be9863401db343bc1d9 : SUCCESS in 6m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/819a584149ba4fb0af55d3827bfc25e5 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66b70ab95f82441ba3a570d3d09404ac : SUCCESS in 5m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3241612f8e64ad294217ea0b0500d92 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/148259b827294906923b01c23db515de : FAILURE in 17m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63802b70326549648764822f5c732010 : SUCCESS in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a6577af10cf4b4ba426c2f43a1d04ce : SUCCESS in 1h 20m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/15dc8741e2ae4475bf2af420184e0246 : SUCCESS in 1h 24m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6ded966ffb9340708234bd890b6fb57b : SUCCESS in 1h 48m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/01bcc49551f44f0198c7868a02743380 : TIMED_OUT in 3h 05m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a62747e52c34cc0b97ebb10935fa43e : SUCCESS in 23m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be00c4e8a3aa463ebc992b87233fc2bc : SUCCESS in 1h 02m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9801fddd52c248ffab656b81eb2f711c : TIMED_OUT in 3h 07m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bec7d606d6e140ba89655a932cb470e2 : SUCCESS in 47m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/769ae8c7a5b84f0e8f90a38d872b71cd : SUCCESS in 1h 23m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e3b4a13fc0114064a19e3c49c0a57bba : SUCCESS in 1h 43m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c1046ccdb8042a5ae8569145acca6a9 : SUCCESS in 1h 50m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14065dbd12bb48e0971f44f72882be4d : SUCCESS in 1h 27m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/736ee05397004b3fa0b5c05ccc3c2fd7 : SUCCESS in 1h 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e6bfd8ccd5a9598add012d6d5b528dccd18e32fb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 21:41:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a966c26e42a92eee861419bf363c8cf97c8d2ba6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-28 12:53:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c81bc751d7414525fe074813c5910220a4d03e93","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 12:53:50.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d5516f0e0ca52097de76a667e5206ff3cdcb80f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 12:54:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93bf22702fd10e13d01345a048cdf0e951ba9841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:58:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c98a2dbc4a0b4ad8b050eabec0a13d51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b6a48aee374431bef8692cf43a4349 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/257f2f9a5b7e4d869885c6d0743f263e : FAILURE in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76eec23bdfe64408b0776d4580ed18b0 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb70b22ed784048a19619204376c229 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/3961afe7b1b0445696f4510aac082fe7 : SUCCESS in 26m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd4d187065a24b5c8d08c3fb69fcdd8c : SUCCESS in 52m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53be0bb9921641ac99c6e557933be8dc : SUCCESS in 1h 27m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92c21606e72245dba97950bb54ced0b6 : SUCCESS in 55m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8dec8203fc54927bba8a6a9bd6f5a7c : SUCCESS in 10m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd2ac9a0bfb140b9bfe5cfd9fa8ad1ff : SUCCESS in 1h 02m 44s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/466a8e31ee50484294be3b2862d8f68f : SUCCESS in 2h 04m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c81316d1ad140dc955f12dc0f08cfbf : SUCCESS in 1h 24m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f055ac8d664d4d39b86727dd4cd21edc : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":2},{"id":"11392ef75220dd8aa485e60dfa2c8f955df655a2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-28 15:17:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48cd10770e00dba079f43745530ec7ce7e6c3803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 15:33:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5f21edcaa7f448c85eed57f7c70aab6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13d0fd45c6544fddb793cf71575555b4 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88fe592f746b4495a61d618d58f6e12e : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dfa2d523a79a6d4930a90b34ed9cc36fe987194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 17:59:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda52bb0d74d43f199dec2c14e256021\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983aed84aa6f4af990b67862624817ba : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a3fa50fe0948c1a05b426252fec794 : SUCCESS in 12m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fdffd3c8aee4c5ab01a6dede65232be : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0378ee107f1444f496e64c19255f097f : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c4696a57050149dcaaba1f1014ce391f : SUCCESS in 52m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b806874075644de898d3be6afc59d3e1 : SUCCESS in 57m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d6d0fba9eef4da1b4e9d5972704e8fd : SUCCESS in 1h 25m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9edf5d4bb3d84930ac1f163725df9bff : SUCCESS in 52m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b76971647cae4462a3e024f08f72da17 : SUCCESS in 9m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5bec7e56b20f4a07acac57ff860cedca : SUCCESS in 56m 33s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/617b09ed95c240d398abe63ba55e4f55 : SUCCESS in 1h 44m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c066418a86b42648728656b6b62a0f9 : SUCCESS in 1h 23m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c5a928839f64148ac076e2e9d36e588 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"04ed0843b137eb7a734a6a11f9b791a8c7c62e23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 17:59:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44ece534530abc89ad8822c1ddab94c8bcfe2cc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:00:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac3b2abbedd244be85eefd3616d5631a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94f0c3af5cc4c54af2f36290cf9c228 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cfcbc03c6db4eaa9d2c85fc78a34705 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3074be4c3b7547dc9422f53542105e41724228b0","revisions":{"355bcab1950f754bdc2775a738cce8e859d1366c":{"kind":"REWORK","_number":1,"created":"2025-03-26 14:48:41.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/14/945614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/945614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/945614/1"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-01-06 15:38:53.000000000","tz":60},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-26 14:48:41.000000000","tz":0},"subject":"Implements cgroupsv2","message":"Implements cgroupsv2\n\nCurrently, Cinder only does cgroups v1. Let\u0027s use the cgroups\nv2 command line, when it\u0027s available in /sys. This avoids\nhaving to boot with compatibility kernel command line options:\n\nsystemd.unified_cgroup_hierarchy\u003dfalse\nsystemd.legacy_systemd_cgroup_controller\u003dfalse\n\nand just work as expected, provided cgroup-tools \u003e\u003d 2.0.0 is\ninstalled in the system.\n\nChange-Id: Ifdfcd480b72727ec182d5a6954c706f365247edc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/355bcab1950f754bdc2775a738cce8e859d1366c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/355bcab1950f754bdc2775a738cce8e859d1366c"}]},"branch":"refs/heads/stable/2025.1"},"3074be4c3b7547dc9422f53542105e41724228b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-26 14:49:01.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/14/945614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/945614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/945614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/945614/2"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-01-06 15:38:53.000000000","tz":60},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-26 14:49:01.000000000","tz":0},"subject":"Implements cgroupsv2","message":"Implements cgroupsv2\n\nCurrently, Cinder only does cgroups v1. Let\u0027s use the cgroups\nv2 command line, when it\u0027s available in /sys. This avoids\nhaving to boot with compatibility kernel command line options:\n\nsystemd.unified_cgroup_hierarchy\u003dfalse\nsystemd.legacy_systemd_cgroup_controller\u003dfalse\n\nand just work as expected, provided cgroup-tools \u003e\u003d 2.0.0 is\ninstalled in the system.\n\nChange-Id: Ifdfcd480b72727ec182d5a6954c706f365247edc\n(cherry picked from commit d41586a115aabf01a756f6ee6ef58a813f87b546)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3074be4c3b7547dc9422f53542105e41724228b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3074be4c3b7547dc9422f53542105e41724228b0"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
