)]}'
{"id":"openstack%2Fcinder~945616","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I32b505aa69c71b62e7e3a52d65d38165d34e97d8","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-03-27 13:46:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32b505aa69c71b62e7e3a52d65d38165d34e97d8","subject":"RBD: Fix upload volume with different format","status":"MERGED","created":"2025-03-26 14:53:53.000000000","updated":"2025-03-28 04:26:57.000000000","submitted":"2025-03-27 13:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945616","cherry_pick_of_change":938265,"cherry_pick_of_patch_set":2,"meta_rev_id":"5299a16bdf31ce83bb1ea292c411fcd83b64a79e","_number":945616,"virtual_id_number":945616,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 13:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2025-03-27 12:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-26 16:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2025-03-27 12:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 14:53:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-03-26 14:54:16.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-26 14:55:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-26 15:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 16:36:29.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-03-26 17:36:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-26 17:43:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 18:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 23:30:24.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 00:21:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-27 12:07:49.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-03-27 22:42:58.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"8647e73d64b12053f9b473d909a323379a1a0f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-26 14:53:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"18cc30418222aa2a1f4d797a4231e5aa02bdd795","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:54:16.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1587/","accounts_in_message":[],"_revision_number":1},{"id":"9280baf41e8190429614d2fe47840a930ac33542","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:54:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945616/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945616/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"2a779280c16241006c426a64dea297059d75f85e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 14:55:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e14f708bf759cbfb036e59a707c35421e1e10533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-26 14:55:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"28a5e3d3647df0efc8a166216679d0266d933eb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:55:26.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1588/","accounts_in_message":[],"_revision_number":2},{"id":"3f1614dfe123e112f9d8398c256cbeecfadd35d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 14:55:28.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945616/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945616/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"f812142fd2a3c04136d77afdf4af64e9b29d8670","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 14:56:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c30961ead36cbd3f7d82d5b3755ec02d08b788d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 15:30:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab3150b8483441c0bcfc50960b27f824\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/304c667b61e947c687887edfa8879eb2 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e5feeec8d54432184780d4cddf2bced : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"132453f5f7c43f21194ca26ee6d210d57e32394b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-26 16:36:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fb7407cd851b2ea99a3439b055d49d1725393fe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 17:36:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945616 : FAILURE in 39m 54s","accounts_in_message":[],"_revision_number":2},{"id":"255e933342b1c2a9f2df14195c59110b3a2b25fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 17:43:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945616 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3d49118e2d1c7f843c7a7923f36a4115f31c7775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 18:33:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aa53e6265904e29b53baeef8315a308\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a8918df46d428f9319c52ba962dde5 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51b12d69d09d459487652fd78030bf1b : SUCCESS in 14m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6df95188b3e04d4cab58c698171dd0bc : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f352f1e7674e8b8aca7a8b96a6cb61 : SUCCESS in 15m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/60d8ea28874f430698af8a22483633c0 : SUCCESS in 1h 55m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b44db79bb7cc425eb088a61d75ec5b42 : TIMED_OUT in 2h 03m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4074d7c50c6f4b6e9df338f004613afe : SUCCESS in 32m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99cf5313fb3a4300bcac060a0ecb9680 : SUCCESS in 36m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1d4c4607fa840c6b94cc1a444e020fd : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99f5d8f1b350463b922a5501cf103dec : SUCCESS in 7m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90dcdf8e98c94b1587c84c4ebbae7a3d : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ed2a80000ece411d9d8b48e1ea8415c7 : TIMED_OUT in 30m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a31bde9ca25241e19d6526acfd8fc349 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/511a2207c8cf4688bd27b9683a71cc3a : FAILURE in 36m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3bc732cc5424e938c7fab0993e95376 : SUCCESS in 4m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5fe95b00b4464847a6cfe4552c741b21 : TIMED_OUT in 3h 35m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e116c85fa524a8bae20343b31f7dbe4 : SUCCESS in 2h 21m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69cb43c60fa544b1ac007ccfa69abeab : SUCCESS in 2h 26m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b8b4e12583b7428284cdcd0534c57964 : SUCCESS in 1h 51m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/68669d3af13445829fd572bacbeacf28 : SUCCESS in 30m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06a0f0d0021f4027a6cc968b22a105d6 : SUCCESS in 1h 01m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a30c40976b38407fb38254870fd6cc4b : SUCCESS in 1h 40m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/771bf05458684e39885a7e2a815a91b4 : SUCCESS in 46m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c10c52f7bf1a4f2bb6681aff96993015 : SUCCESS in 2h 18m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/386fc4e073f74ec0b4250def5d3399de : SUCCESS in 2h 07m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d3619ea00a843caa9c43e2add470ef0 : SUCCESS in 58m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36529d1e2aa5450b82ce22dd327ad69b : SUCCESS in 1h 55m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/946ee93120dd48ecb3708ac7eaf18ad5 : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1f69a04e962b30f304b0fe1bb49510c4d4312c07","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-26 19:10:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e5831be3dab12a67d4064b5269ebf52b3b15b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:32:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce8cc97d17a4b6d81cfeed502546c57\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8d0d5e4e74c404bb1aa15aaf6b497c7 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97bff1ff92584a4bba5dbb638014ccb0 : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"657f9f9acf699503939c56a89f7235f30886f1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 21:19:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/127bf1ab0ada42dd8696cb762fa2fe60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a93dd8679384f75b963f88741cfff5f : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42be0ed70ba04b03a1095f27670f0c5c : SUCCESS in 8m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9ac2f29a5994b478caf9cbabf06418c : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e93a498b424aa09abc0781bb08aa1c : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/32089ea8c4f54d16ba9b0bf947a6a377 : SUCCESS in 54m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/106d1ac56be14e2ba25b20ad294aed19 : SUCCESS in 1h 29m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38c626f2d3694a8c94cbdab370c46a9c : SUCCESS in 52m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad35b41854424854b049157cb39a8abd : SUCCESS in 22m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c408772fb7394e129f470be03e98f7b9 : SUCCESS in 19m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/52cd3b2e6a13422f9a17a07e1f7e5187 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfa8365b72264244bcf64314a3374263 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fbf33070f7742679936bc556a3c577c : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d53e3625a2d443ab1856607151773c9 : SUCCESS in 10m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f08f535401243d4b911fe1b0d90b400 : FAILURE in 29m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/164dba7859384d1b9f6b02886e52c0ae : SUCCESS in 4m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f45f94dbd7840dcaeb24f7c09f42fa6 : SUCCESS in 1h 50m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6275994ab4e04eaca76827e12e5f8dc4 : SUCCESS in 2h 08m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf25cc00776c4913b3f75da21e5b8ce5 : SUCCESS in 2h 07m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/99fa8c21fa9541afb50a18efe327c159 : SUCCESS in 1h 54m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e56e10ccd244d918de6f4a145f032ce : SUCCESS in 21m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7913ec8e70494a7c897c53ef1a13ccd9 : SUCCESS in 58m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6a6cf42b969f4c22831e8d79799743eb : SUCCESS in 1h 26m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/438f6e67951a4f628e68bd0db0dc5a6f : SUCCESS in 27m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/595da86711a548298c6987b486ae6345 : SUCCESS in 1h 32m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6d41b7832ccb4026a63e68e736ed0b4f : SUCCESS in 1h 16m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/696d3d937f224a2daa42460ffb0a2233 : SUCCESS in 1h 20m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f7be1585ddc40cfa57549ace241ea95 : SUCCESS in 1h 38m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccfb5d42993d400dadbe5da7afe61374 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0b49d96e0c23ed34d763febf612308280fc577e6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-26 23:30:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"de62eaa074d0440f14347db52b370d64db365255","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 00:21:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"fb5eacc9b82b7cd848e325e740eb9be2eb658781","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 00:47:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9b90deff29d9dc4bd0fd23adba8cb84fb605c310","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-03-27 12:07:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9f24ec6569bbb479280f3ac925deeb2fdd86f76d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:08:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1770008f12596ad58aa7aa6a89814d2878816e29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:46:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2701ca1a1f8844e6be6e494b52cfc2c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549d3e28b88f4ec38982dfc59a6f1d79 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32e4064038e940d18156ac3522b9377e : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a32fd80050b24384868c8a5bc09f0764 : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464b677e59e34e9e8acbdd031b626220 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a4c9be7948394999864e75484c3e2516 : SUCCESS in 1h 01m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/767adf47c76e4ebfb487ace137e5003e : SUCCESS in 54m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f5ea6827f3243518f5b20867c941e71 : SUCCESS in 1h 33m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8df1e385c2c34874bc1489cf6100a191 : SUCCESS in 1h 02m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995dee5801914147bbeab6218b789854 : SUCCESS in 18m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e4ce0d972e941b6a87cbc8dd1c52873 : SUCCESS in 44m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c91ea578c2e746dc8df398b2f85d52f4 : SUCCESS in 58m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f38ab7e5647a48aaa14f78cfa285ffde : SUCCESS in 1h 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cdad20f1d6649139847323511234eee : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ddab48426c64b17c5df99fded65190cdfe879513","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:46:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2602c5574c2272520a5e9a6ac31a23aafc2a1309","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:48:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1867bf5194a4e0e9d4b622ad9fd61e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f422c263d734297995c102390ebbd07 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f933385939904115b597567cb352f283 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2},{"id":"2f0c5ce8660e27ba44fb1edccda2e77d00aee2eb","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-27 22:42:58.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5299a16bdf31ce83bb1ea292c411fcd83b64a79e","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 04:26:57.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/7fe69efc22f140cba96e80c6b898308a\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/044563e5e5014b758f382950a534b580 : FAILURE in 1h 10m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb8ad77e60098c49f2937b13463e695103587e1c","revisions":{"49a5f8ad6439d7ba08dceef87127a1e8d99f60ef":{"kind":"REWORK","_number":1,"created":"2025-03-26 14:53:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/16/945616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/945616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/945616/1"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-26 14:50:19.000000000","tz":330},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-03-26 14:53:53.000000000","tz":0},"subject":"RBD: Fix upload volume with different format","message":"RBD: Fix upload volume with different format\n\nWith change[1], we introduced an optimized path to upload an RBD\nvolume to image. However, this does not work well when the image\nformat is not \u0027raw\u0027 and conversion is required.\n\nCurrently there are 2 places where we need handling of the\nRBDVolumeIOWrapper object:\n1. format inspector\n2. qemu-img commands\n\nWhile testing, I found out the following issues that needs to be\naddressed to fix the upload path with conversion:\n\n1. Passing RBDVolumeIOWrapper to format inspector\nWe fail when calling privsep since RPC cannot serialize the\nRBDVolumeIOWrapper object and fails with the following error\n\nERROR oslo_messaging.rpc.server TypeError: can not serialize \u0027RBDVolumeIOWrapper\u0027 object\n\n2. Handling in format inspector\nEither we need to pass the volume_fd var to several methods in\nformat inspector code or introduce a condition to find out if\nwe get volume path or RBDVolumeIOWrapper and open it if it\u0027s a path\n\n3. Handling in qemu-img commands\nWe need to use rbd:{pool-name}/{image-name} format instead of file path when\nexecuting qemu-img commands like convert[2] otherwise we will fail with the\nbelow error.\n\nStderr: \"qemu-img: Could not open \u0027\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x781f747a6aa0\u003e\u0027:\nCould not open \u0027\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x781f747a6aa0\u003e\u0027: No such file or directory\\n\"\n\nNot to mention passing the volume_fd object to all the qemu related methods\nand handling cases when volume_path is none and volume_fd contains the\nRBD IO object.\n\nEven after fixing all the issues, it\u0027s still not sure that the optimized\npath offers better performance compared to the traditional workflow as\nqemu-img will use librbd to read the volume file and convert it to a local\npath.\n\nDue to the above reasons, we only use the optimized path when the image\nformat is same as volume format (i.e. raw) and container format is not\n\u0027compressed\u0027.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/928024\n[2] https://docs.ceph.com/en/latest/rbd/qemu-rbd/#running-qemu-with-rbd\n\nCloses-Bug: #2092534\nChange-Id: I32b505aa69c71b62e7e3a52d65d38165d34e97d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a5f8ad6439d7ba08dceef87127a1e8d99f60ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a5f8ad6439d7ba08dceef87127a1e8d99f60ef"}]},"branch":"refs/heads/stable/2025.1"},"bb8ad77e60098c49f2937b13463e695103587e1c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-26 14:55:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/16/945616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/945616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/945616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/945616/2"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-26 14:50:19.000000000","tz":330},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-03-26 14:55:17.000000000","tz":0},"subject":"RBD: Fix upload volume with different format","message":"RBD: Fix upload volume with different format\n\nWith change[1], we introduced an optimized path to upload an RBD\nvolume to image. However, this does not work well when the image\nformat is not \u0027raw\u0027 and conversion is required.\n\nCurrently there are 2 places where we need handling of the\nRBDVolumeIOWrapper object:\n1. format inspector\n2. qemu-img commands\n\nWhile testing, I found out the following issues that needs to be\naddressed to fix the upload path with conversion:\n\n1. Passing RBDVolumeIOWrapper to format inspector\nWe fail when calling privsep since RPC cannot serialize the\nRBDVolumeIOWrapper object and fails with the following error\n\nERROR oslo_messaging.rpc.server TypeError: can not serialize \u0027RBDVolumeIOWrapper\u0027 object\n\n2. Handling in format inspector\nEither we need to pass the volume_fd var to several methods in\nformat inspector code or introduce a condition to find out if\nwe get volume path or RBDVolumeIOWrapper and open it if it\u0027s a path\n\n3. Handling in qemu-img commands\nWe need to use rbd:{pool-name}/{image-name} format instead of file path when\nexecuting qemu-img commands like convert[2] otherwise we will fail with the\nbelow error.\n\nStderr: \"qemu-img: Could not open \u0027\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x781f747a6aa0\u003e\u0027:\nCould not open \u0027\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x781f747a6aa0\u003e\u0027: No such file or directory\\n\"\n\nNot to mention passing the volume_fd object to all the qemu related methods\nand handling cases when volume_path is none and volume_fd contains the\nRBD IO object.\n\nEven after fixing all the issues, it\u0027s still not sure that the optimized\npath offers better performance compared to the traditional workflow as\nqemu-img will use librbd to read the volume file and convert it to a local\npath.\n\nDue to the above reasons, we only use the optimized path when the image\nformat is same as volume format (i.e. raw) and container format is not\n\u0027compressed\u0027.\n\n[1] https://review.opendev.org/c/openstack/cinder/+/928024\n[2] https://docs.ceph.com/en/latest/rbd/qemu-rbd/#running-qemu-with-rbd\n\nCloses-Bug: #2092534\nChange-Id: I32b505aa69c71b62e7e3a52d65d38165d34e97d8\n(cherry picked from commit 7d320764980a0e33a9abad14db41a4cdee71d57a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb8ad77e60098c49f2937b13463e695103587e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb8ad77e60098c49f2937b13463e695103587e1c"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
