)]}'
{"id":"openstack%2Fcinder~945617","triplet_id":"openstack%2Fcinder~stable%2F2025.1~Ic4bf44c320ad53b514178ecd4d5f57f037169bfe","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-03-27 13:46:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4bf44c320ad53b514178ecd4d5f57f037169bfe","subject":"Fix reimage with snapshot backed image","status":"MERGED","created":"2025-03-26 15:00:23.000000000","updated":"2025-03-28 06:12:14.000000000","submitted":"2025-03-27 13:46:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945617","meta_rev_id":"2213cf2a29b6014b424f33b7d2f3daa60b0dcca3","_number":945617,"virtual_id_number":945617,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 13:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"recommended":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":1,"date":"2025-03-27 05:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":2,"date":"2025-03-27 12:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-26 16:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2025-03-27 12:09:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 15:00:23.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-03-26 15:00:31.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-26 15:01:21.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-26 15:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 16:38:10.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-03-26 17:41:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 18:20:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-26 18:34:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 20:43:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-27 02:14:51.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 05:39:26.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2025-03-27 12:09:08.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-03-28 01:00:28.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"9e29cae2988342b9df5002b5ef0d0fe573b7993b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-26 15:00:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e4f7815f62f0ac1a5c5c15554c199f3f1f0302","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 15:00:31.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1589/","accounts_in_message":[],"_revision_number":1},{"id":"3a62ab94133a9c83b2f5aba0f10b02509da64f0c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 15:00:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945617/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945617/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"ba958fa68e64fd988d06d0412c4108a985fe0420","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 15:01:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d5064d985ab5c6f18e7dd35cb289acc79861b4c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 15:43:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1951845819a4444a36cceedacd41a99\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5930b75eb3384f208845281fad3b0eab : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a4931546b474f22860a25b21aedef89 : SUCCESS in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edbb0846fcb0d1792c1f6c576a34d36b2b1fe83a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-26 16:38:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f95bccf157d11934edc940c3a29f59557f96d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 17:41:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0bc380967447ddb628a20d95d5d337\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51465a6c5d29484ca6461cdec9cabf1a : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a737305a95d45d2bd3d50d692bf23db : SUCCESS in 14m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d8b2da130e84a8388f19b8eb7d9bc48 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8464176eab1d4b0daf059eb56473a97a : SUCCESS in 15m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/2c6e1f5927ab4ac9a59de654d67bcf1e : SUCCESS in 55m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f0228c35ce14c2fa8bb613818cbb436 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36c9135034724ac3bfb8c8d7cc489635 : SUCCESS in 1h 00m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f93a33bf3c14f56aea3e2690d103c8e : SUCCESS in 18m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5d2254ce8d64fd28a8530993ab66271 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ad5ee8dfce4745df988fb1823cee9a11 : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c6f9b51fa10947eeb9ee3ae3df30f7a3 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1604f362a6ad421aaa293f129764970d : SUCCESS in 6m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7e32784094344748818ea03ef10db40 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2a3f0f01088423f910918c88886a596 : FAILURE in 33m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2130ba23983415e81c90c9a23c974bb : SUCCESS in 29m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c84bf954e6549a58ed4381003df2e15 : SUCCESS in 1h 14m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eea7384cd12b48c48eb110634c4ca5cb : SUCCESS in 2h 34m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f353889ff875452b95f4d2b6a840d838 : SUCCESS in 2h 36m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/06d0a2df4ed94ac78333880d59a2129c : SUCCESS in 1h 43m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/decdd53d1a3a4d899383051afbb6986f : SUCCESS in 1h 25m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/920ea2e1cdf24fc8bf822bf51bc456b8 : SUCCESS in 1h 16m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93154a5c6d724cf2808669f0b7b9a53a : SUCCESS in 1h 48m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43c20390f932497fa9dfa5ec918e5ac9 : SUCCESS in 51m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/117b36d760354e7cb05b1eed5ff269a4 : SUCCESS in 1h 19m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d49b44f413040f297c3e5ff83bcd6de : SUCCESS in 1h 43m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3decedd27ec347068ed64f5719c6912b : SUCCESS in 1h 34m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810e1768fa2f402b82540a68bf6653d6 : SUCCESS in 1h 35m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/063125d9d94a4eeaace72aca9a591030 : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":1},{"id":"18003b02288f8e7477d8e6587769e19f435a8691","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 18:20:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945617 : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a60bd055d89eee4086958efbf5ffa99a3a958722","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 18:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945617 : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"06861cac854e13d40e520ec9de884d1cefb6e2f7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 20:43:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"92bee497b4e4eff18510bfcf96b4d33786a0bb30","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 02:14:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"049229a2d0802b5397b9cf4e3f162a22774f61a2","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2025-03-27 05:39:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1890d0e3e487bee53f777acc5e68e8a08622a840","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-03-27 12:09:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68db17cf55914ed5ac50e7b1c5e1793865562ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:10:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dae2e507cadfd21c0972ea96e48536971b0000e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:46:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6797aecbe34f1aba8e5a831823adbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7e8129cb664573b25035055369a2bc : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1821ba589bf463db00daec095b90303 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/683a72cfdb9944219970313319221742 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca3432a00f346cc99f6fc627f8a3362 : SUCCESS in 14m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad1239ffc9e4daf947735ecb1b80dba : SUCCESS in 53m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cd19f32d08742a8b410456d8a5c37c2 : SUCCESS in 50m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b877b810c0724e0e9f9f2650dcccd8d4 : SUCCESS in 1h 31m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db24e9af322e4ae1b79e3d151a83ceff : SUCCESS in 32m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4ea38e5853b4e20888ed9741fb82942 : SUCCESS in 10m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37fa4cc7e9f4456fac878a5182d081ba : SUCCESS in 43m 57s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc0076dc61b8495082ca6ba5f0e1442f : SUCCESS in 1h 21m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fef677701db54f96ab6088d1bdf096b5 : SUCCESS in 1h 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7522a98be5354518ba28fec017f62a16 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"379fb1d6b1959697877a92eecc41c0dd1a735d58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:46:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee5b7369c3e90c461f2d7cdba17ea1c9c7b41e33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:48:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1750152a677849719a46739a24c8ea1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a749e233e19f4fccb9e12a915b88a9eb : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d51608c7c86949e78151ca46ed73d003 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"6d032416c22eb0bdff245cce3cdf8348a615d1c9","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 01:00:28.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2213cf2a29b6014b424f33b7d2f3daa60b0dcca3","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 06:12:14.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/6aaed5b50f5d4d2eb7913a5ed19ffadf\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/a435eebe74fa4c008c51f581f2ce3e88 : FAILURE in 1h 45m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25a2970124d8be372ae366ed8bb9bbacc604271f","revisions":{"25a2970124d8be372ae366ed8bb9bbacc604271f":{"kind":"REWORK","_number":1,"created":"2025-03-26 15:00:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/17/945617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/945617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/945617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/945617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/945617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/945617/1"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-04-09 03:24:55.000000000","tz":330},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-26 14:59:48.000000000","tz":-240},"subject":"Fix reimage with snapshot backed image","message":"Fix reimage with snapshot backed image\n\nWhen we create a snapshot of a server that is volume-backed,\nglance creates a zero sized metadata entry of the image which\nis backed by the volume snapshot.\nWhen we try to reimage a volume with that given image, we need\nto fetch the volume snapshot details from the image metadata.\nNow to reimage, we need to create a new volume from snapshot and\ncopy the data from our snapshot-volume to our original volume\nfor which we use the generic revert to snapshot mechanism as\nit performs the same steps.\n\nCloses-Bug: #2062539\n\nChange-Id: Ic4bf44c320ad53b514178ecd4d5f57f037169bfe\n(cherry picked from commit ae22195df7d674ba017f5e301868522ccb21c5b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25a2970124d8be372ae366ed8bb9bbacc604271f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25a2970124d8be372ae366ed8bb9bbacc604271f"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
