)]}'
{"id":"openstack%2Fcinder~945631","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e","project":"openstack/cinder","branch":"stable/2025.1","topic":"bug_2101859-lacp-stable/2025.1","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-03-27 21:49:21.000000000","reason":"Change was submitted"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-03-26 16:22:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e","subject":"[Pure Storage] Fix issue with LACP ports not being identified","status":"MERGED","created":"2025-03-26 16:19:57.000000000","updated":"2025-03-28 03:16:31.000000000","submitted":"2025-03-27 21:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945631-bug_2101859-lacp-stable/2025.1","cherry_pick_of_change":944045,"cherry_pick_of_patch_set":4,"meta_rev_id":"ed6ba6fb28622d924551b7a4027c0bd678a12332","_number":945631,"virtual_id_number":945631,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 21:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-03-27 01:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-27 05:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-27 05:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 16:20:26.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-26 16:21:06.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-26 17:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 19:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 19:49:07.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-26 20:16:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 21:25:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-27 01:13:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-03-27 04:54:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 05:09:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-03-27 12:42:43.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"3f09f457f74ba495f5d7fdff71fa4d973f87feba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-26 16:19:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ccfec4255edae029e66ce0202a50fdc362a0b975","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:20:26.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1591/","accounts_in_message":[],"_revision_number":1},{"id":"233058c1dc91791daa420bef254a624a684ebce3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:20:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"9ffaee76f7f213cd46173150a862590213040605","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 16:21:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"38f5e48f79314655f3789f0ee2a2d2af8e11c48d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-26 16:22:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8fe868696960216c1904f1b239aa58357d8b349c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-26 16:22:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"553b483d739e33601802c9bc357d5613b3e74a09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:22:56.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1592/","accounts_in_message":[],"_revision_number":2},{"id":"f9a0800547cc1d2fc3fa9b881cd455d21bac01dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:22:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"15ed2984864e47f540c3a36dc09499ddea9d9213","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:23:06.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1593/","accounts_in_message":[],"_revision_number":3},{"id":"7d41980c346ebaba62c1846a5cacb310d1d04f62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-26 16:23:07.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945631/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"df2e960c597eda1717974db51359d47947ec6ab4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-26 16:24:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"f97ce0ffeafe343a0fd8fb2464d4fac16e89b98c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 17:48:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4580e06ea044e1a43e4a5aa5a1ca2d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb8a2ae2d3a547a19c828c925c3c1a46 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/840178c518e448b9a913002c7bc36903 : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8cfdb1fa194d5dbf4bfb1e80f88927a843c75427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:24:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2e4e1f982ad4748b48f33ec1f8b4ae5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a464ff558c146e986d6eeac30c6b060 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb2e2720ab6347afa10b1a2aa40878f5 : SUCCESS in 7m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f58054196394b8d8252d1970ce52993 : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b585575fc8a94c8a83f3f7bbc81c0265 : SUCCESS in 14m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/bce297ba3db14f308e0feed15ce8a20f : SUCCESS in 53m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f808a78898bb48d5a8e5c3fa70858786 : SUCCESS in 1h 27m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfeae9d77ee343a18372c091e3f10d3a : SUCCESS in 1h 33m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c18d71d37bcf4856a7fef23283f0858f : SUCCESS in 18m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e1a84c30333485ba72c8db203061ff6 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/688c46f7eaff4947ae9096f2d9bf8eb7 : SUCCESS in 7m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/750da9a51edf43888b0087d86bff20f9 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4243ccbe75634021a547ebe69f8f96cd : SUCCESS in 17m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b2228e65e55c485da1bc6f8520ff692c : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/810b4a1a322d47d7bce32e161966a962 : FAILURE in 31m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f1ba38de62b4443e8ea741ba172adc55 : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/06438a311e7d401fb561a6cb0ee5a673 : SUCCESS in 1h 56m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7165f4b44d5142e39ed22a074e6d1308 : FAILURE in 2h 29m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0211e7106d5141aeb6e1e676ec7b7913 : SUCCESS in 2h 17m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/062775dcecd84299b0e78a0501f0d97b : SUCCESS in 1h 48m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/44a438d8ae90464c99b47170a3be8f06 : SUCCESS in 35m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92960a58791544ee94ce0c35704d8a50 : SUCCESS in 1h 00m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/253b2419e93641e498c809f4479c36b0 : SUCCESS in 1h 30m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f42b408c9f074e60a9118d97c3eb1097 : SUCCESS in 45m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25f08366ad694abcb62f8bdc98dd1cab : SUCCESS in 1h 20m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/785171277f9e4067b09bda598c82ab2f : SUCCESS in 1h 35m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b301b267d1534d6cb4bd5f76d71a234b : SUCCESS in 52m 02s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde87c19b9584831b7b211e823a1ed82 : SUCCESS in 1h 36m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44e71cde73734937b3ecccfd45724f80 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b6c454e4073fbf8704093afd1eb70e34265db943","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-26 19:49:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945631 : FAILURE in 41m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fa4462a21f1c555851587c76249b04e785bed13d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-26 20:16:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945631 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":3},{"id":"729621324f81cd9498b437352aa1fdc68bcd80e8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-26 21:25:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4cb39f276b177c5f97b0fef45037a7e975a2fe81","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 01:13:59.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"19a26d0a961523e2a6219e24470bf14281ca68fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 04:54:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e1c19991a23a7366697513e468dd59c56c395125","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-27 05:09:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37cc3282e707eba73b09da4114bc0edb53389fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 05:10:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"12758f5890729acf514c34f413691179e9991a5b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 06:41:27.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"002f5438b9904c2f4301b9c3f8e8c5e65fd88135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 06:55:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3c7bfbc0af74ede83387af74bdae4ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d5c22a3f864e96a59aae88708f8d73 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27bf8571bb3468590d43669afcdf269 : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50f8891960f5450a942c37ef8251b72c : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee534c58a674142b2649bc7ae73ab59 : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a3aa1bc6408744febbf4660b37b9d95f : SUCCESS in 1h 04m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a03ea7017161405d95e734ac3c5d5ed1 : SUCCESS in 52m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd0e32d841444881bdb87879b7bbd97f : SUCCESS in 1h 39m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37624cb58ed8468b8a779b22924d5ed3 : SUCCESS in 57m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86bafdc86ec44f839017388566461624 : SUCCESS in 9m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4da2f8df1b94b59b1475e5a5186f37d : SUCCESS in 1h 06m 59s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da6f3d1138b247a9be532cb5201b1e50 : FAILURE in 1h 09m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/203c4194e2e94138b83e9eb37bc18adc : SUCCESS in 1h 22m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/443930ea9e044d928f6cca406e8655ab : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f7f8c93e26aa16188004a72038d713f6a0a9cee3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 07:45:59.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"90428916b75308524cced9757e3bcb1cfa1eb246","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 11:13:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8034bf5ee704fd21ce3f6d50f03a2da2697b944f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:41:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23aafdc82e654c80b9ba9dbb1b5bc5cd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eea5147858664fd58485e392fa5fdc8d : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a13100bcbbca4216af5af5881d81aac0 : SUCCESS in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4ceeb36da1946ca813bcd57be7fffe0fbf62206","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-27 12:42:43.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21709721c56c0583fb60d0089aa5dc4dce398632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:53:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e75a07d66c4442f59dacd7c914a9b829\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a171d8c4623e444280b8f2f46f388dec : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c19441ed4934bc38f586704275c2035 : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc5498f6dd53409cac7a94c233b64240 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d29a8b8f7241959a1922400e8bea36 : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5ee116e929a847e39edefbcf18747d0a : SUCCESS in 57m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7271135e8e9242a9bd669258cdf8ac16 : SUCCESS in 48m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d943f5c9456643d7813100cd4f612501 : SUCCESS in 1h 20m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f9ff40498244fecb3547fab72fcbab6 : SUCCESS in 1h 00m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4eb2c53e2094b96912182b06044f1d3 : SUCCESS in 19m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53ad73576d6a4121a7e1368c9eba325c : SUCCESS in 23m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ce4bc61f22e4dffa5b724c775ee8633 : SUCCESS in 6m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f39eab9743e54357ad849afaac80049f : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ac4afa8a9894fccb815c50a88cc0fd0 : SUCCESS in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/820e283ae663492ebc79ff3ce17b1114 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51e5b1113af6490c8d6422b6d35fac94 : FAILURE in 33m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a7f46c8ef454d95bbb194bcd20d2226 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71c5b7f00de845b28161842caecb1ac9 : FAILURE in 2h 07m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb59e8f38c874a1cb4e4f5145992b26f : FAILURE in 2h 27m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8c85537b4f2f4f6bba220f33eb97835b : SUCCESS in 1h 48m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/72a4221f0d03468d8f182a8ce45256fc : SUCCESS in 1h 52m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/95fa513159ce44d982f2a0ced2af49de : SUCCESS in 29m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ea95111db764bdc8919970550fb8fa4 : SUCCESS in 1h 07m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ccd5bdbd55648eeb1373d503d7d36e6 : SUCCESS in 1h 19m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c8d21f41317449bdb5667198ec846492 : SUCCESS in 35m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/10ef8ffa510747b48de0f970e43cffa5 : SUCCESS in 1h 32m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/46fbb0e0caf844d4a3044c99f90f1952 : SUCCESS in 1h 44m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e39b13bc78f94f12a069ff48b2116223 : SUCCESS in 2h 00m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1defb666c6ba45d6aaf98b938844bec6 : SUCCESS in 1h 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d276e48aad314602a75283e348efd45b : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":3},{"id":"8a78228e7baef8b1d2041b427387b1c830d45469","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-27 14:06:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a759e200f07f1cc597deb0ddf3454975e1999921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:23:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca02329ec5c74769a6e9f87e50094fc6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23481e665c7d4cdb88dc58f7679f1f57 : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba40eabad60a43bd9ab52731b8c60a0e : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d992754f7bc397d93402d1c9da195e1db10d7f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 16:57:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebd31813f4ba4bd9a51d70f3d5510362\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c873190e4f754d23b6a4ae31f0f0f465 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdda4659fab642ce8f78109f707c0629 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc0139a353f046e799841ffd11be9278 : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d4c83f66294862a6e32aef8b3388fc : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/3b4bf209e2514e15815c8b95822da310 : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a57ae30ba7b74818890ff63b3e8b2880 : SUCCESS in 1h 02m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7adef1789ab4103b8bc5f5497293238 : SUCCESS in 1h 28m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5bac5778cb143119ed0b833a0f168c1 : SUCCESS in 55m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2e4b736c6d44374bb94392a99e28543 : SUCCESS in 24m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9367223da9c54313a0badc16bc7513d2 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5018f71dbc8a434283bc0b1b331703a1 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93d72356ebe7429ab5ae24ec757d0e0f : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be1685e1bf3d4c16acb6b47a42c5a3f9 : SUCCESS in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3c7ae5902ba943a9a1533c6373639361 : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c07dc4e880f3476789f0df31e40e0254 : FAILURE in 33m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c58f0bfc02a418c90edb5f165b8335a : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f6fa064421a4ed789eabff434b2b06b : SUCCESS in 1h 52m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ccf34e835664536b0e6745c227424d0 : SUCCESS in 2h 13m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/86eb27ca1a1b4e09a36a616c8e21ec24 : SUCCESS in 2h 00m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b2c120b253684d62b75a1119851f5775 : SUCCESS in 1h 53m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1f6c897e09e34497837e04f0d4a088ce : SUCCESS in 32m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64b96e5b275d4227b85eb298a5164460 : SUCCESS in 1h 05m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f38b937db1b49d88cbb8200760f0213 : SUCCESS in 1h 49m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fadfe82d50714fabbc385370592a8124 : SUCCESS in 26m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9800d1600dc24c31ae224f34f154389b : POST_FAILURE in 1h 43m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a66b49745c524c0481ee13e46127a985 : SUCCESS in 1h 28m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94a380e6f79344d2a2ed14c5118ee8a5 : SUCCESS in 1h 59m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3ab80f57d2f4687908dc5c01f1db7c8 : SUCCESS in 1h 37m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70649675fce449c98787e3eee866cda5 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":3},{"id":"cede7823801be52963fa1aa3bfc9d6c36c6d41a2","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-27 17:06:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e17df19c8697218ed008c9abc1d7c54ba529b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:47:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db6c09d3cb3940aab45cf7a568f66cad\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9282353e6d4a412b987056d932cda990 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cd264d3c5df4bfaa7c87373380b7228 : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6eb2fc3c8bee97200d09d8eba2fdeefcd5762948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:34:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/652b0ce37cdc4647b061f0efa67b9811\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a04f456d4ce245d289de9af9306d4e94 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c9d6e5b16ce4a0d80009dece7544b20 : SUCCESS in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/126893e2d5564845bad8889044d4aea8 : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cfe0dc9f554a03b41c4544f4fbb096 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/bc2cb7d8a0fb4a80b8ecd63960c7c6b7 : SUCCESS in 54m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fec0a471eda433f9c389299b0ad2a52 : SUCCESS in 1h 04m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1589a714f34443ab8ae523ba42a7bb61 : SUCCESS in 1h 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06489e6b12354ab2aee810c1a8244fa0 : SUCCESS in 59m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9a336a85ca4490da688563d2cc8ae7e : SUCCESS in 19m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/636ec2f4fb794546b3a16123e20f6f5b : FAILURE in 22m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/178d0bdeee0740e9b9e1ff79902600c8 : SUCCESS in 2m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b19fc034abe343a2a1b614cfaaf55f1d : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d7fbf041258456c9a09a7406e0f2f8c : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/00730df5dc66494483ccc7b80c481936 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/458c51acbe814c208a99fab89d8614a6 : FAILURE in 32m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12afb6752052413b9d56e1967fa99e25 : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3cb76a537d9e4e4e9468d5c885390294 : SUCCESS in 1h 57m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/26542ccaebcb48b39af35d35aeede4a9 : FAILURE in 2h 25m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ba8f06ebd5c24cd99aaefa4c0fab1cc8 : SUCCESS in 1h 57m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c9fe1af1d1cc4bd18758c297f02f3f52 : SUCCESS in 2h 07m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ce58b1bc6feb4d178936b6ec2e22840d : SUCCESS in 36m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b670fb1476d84ba2990f4dd4891e8742 : SUCCESS in 1h 03m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ab1e39356c74c81825fbb07fcb2eb6c : SUCCESS in 1h 46m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/66ecfbaa573a4ce88de5e219ec76dbba : SUCCESS in 1h 18m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/273d4953c5f94d53bdfbefd338a12ebf : SUCCESS in 1h 26m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2b1f7220a7da43bcae18ecbd2667a63d : FAILURE in 1h 15m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6bca12163e57406984e2cb26114f3297 : SUCCESS in 1h 57m 42s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a156696031e4b5b982176ed30b3a2af : SUCCESS in 1h 41m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/419bd88dc8624d59ad9f2f08dccd01d6 : SUCCESS in 54m 53s","accounts_in_message":[],"_revision_number":3},{"id":"fda2975738d5a3a255f67e3025a663d4f6f10695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:35:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a83a36637ee0c205ce2c1bd5fd33a615e6e797a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 19:35:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1bcc8e1af241d5c44f8ebaaabd268b8d29aea0d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 21:49:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17754038e3584eca9fdccf66ef8ea915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668398821eff416a87a13d506d4a777d : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/015656b57a6d4e12be3cd63c5b70e75b : SUCCESS in 10m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1954c1dd8581437eb7bc2508bb99a0fe : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d2bc8242a2452581c9c125facc0f8d : SUCCESS in 16m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/61f197c8ad3746059027520f35ef15b8 : SUCCESS in 31m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b48fed06558e45708f6c7c139790a66d : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48d26eb8dcfd4dfab56358eb78128132 : SUCCESS in 1h 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db11d855950c420abc797c166dcdab59 : SUCCESS in 58m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259d4789887141778c1cf28efcfb6409 : SUCCESS in 19m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd0b4976eceb4da9a9a5402c59f7356d : SUCCESS in 1h 07m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efa8c7b9bcac4b64bb1cbf27113f0cac : SUCCESS in 2h 09m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac3a2015a6dc46f4bd077b42be5036d5 : SUCCESS in 1h 24m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/417e7c46a2a641998190dddbba242b77 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7ee2eebb3aa474c7920dc7a7b897cfd7858bd2db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 21:49:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"291813a287c0e7b3a21ac77dbd9ac59fc4e98938","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 21:50:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db13c07a493944a8a0984cd1f1be579a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614435654a9145a7a772f2ac1bc8693e : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bfae1d45e0f42b0a1409239095bf672 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3},{"id":"ed6ba6fb28622d924551b7a4027c0bd678a12332","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 03:16:31.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/14056922f8f649a88bd078c3d75a60b2\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/c664dbc71f6c4e8fac6649ed7a19aa5f : FAILURE in 2h 16m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d5da037e6c012312d12d224df5a19385c29d7c66","revisions":{"e3f3e8e57b2f2b96044157ec72b803338737e4b4":{"kind":"REWORK","_number":1,"created":"2025-03-26 16:19:57.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/31/945631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/945631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/945631/1"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-10 17:31:56.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-26 16:19:57.000000000","tz":0},"subject":"[Pure Storage] Fix issue with LACP ports not being identified","message":"[Pure Storage] Fix issue with LACP ports not being identified\n\nWhen using iSCSI or NVMe as dataplanes target ports are identified\ndirectly from the backend.\n\nWhen an LACP bond is created this was not being correctly identified\nas valid.\n\nThis patch resolves this issue.\n\nCloses-Bug: #2101859\nChange-Id: I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3f3e8e57b2f2b96044157ec72b803338737e4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3f3e8e57b2f2b96044157ec72b803338737e4b4"}]},"branch":"refs/heads/stable/2025.1"},"b22b6a9d12910739a77c5f71391bd8b4c3648f9f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-26 16:22:44.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/31/945631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/945631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/945631/2"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-10 17:31:56.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-26 16:22:44.000000000","tz":0},"subject":"[Pure Storage] Fix issue with LACP ports not being identified","message":"[Pure Storage] Fix issue with LACP ports not being identified\n\nWhen using iSCSI or NVMe as dataplanes target ports are identified\ndirectly from the backend.\n\nWhen an LACP bond is created this was not being correctly identified\nas valid.\n\nThis patch resolves this issue.\n\nCloses-Bug: #2101859\nChange-Id: I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e\n(cheyry picked from commit 5070eeaecf2837c8f5109c63518d87a08f9c40b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b22b6a9d12910739a77c5f71391bd8b4c3648f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b22b6a9d12910739a77c5f71391bd8b4c3648f9f"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"d5da037e6c012312d12d224df5a19385c29d7c66":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-03-26 16:22:54.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/31/945631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/945631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/945631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/945631/3"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-10 17:31:56.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-26 16:22:54.000000000","tz":0},"subject":"[Pure Storage] Fix issue with LACP ports not being identified","message":"[Pure Storage] Fix issue with LACP ports not being identified\n\nWhen using iSCSI or NVMe as dataplanes target ports are identified\ndirectly from the backend.\n\nWhen an LACP bond is created this was not being correctly identified\nas valid.\n\nThis patch resolves this issue.\n\nCloses-Bug: #2101859\nChange-Id: I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e\n(cherry picked from commit 5070eeaecf2837c8f5109c63518d87a08f9c40b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5da037e6c012312d12d224df5a19385c29d7c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5da037e6c012312d12d224df5a19385c29d7c66"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
