)]}'
{"id":"openstack%2Fcinder~945666","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea","project":"openstack/cinder","branch":"stable/2025.1","topic":"bug/1886543-stable/2025.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-03-28 14:39:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea","subject":"Driver assisted migration on retype when it\u0027s safe","status":"MERGED","created":"2025-03-27 01:20:29.000000000","updated":"2025-03-28 14:41:37.000000000","submitted":"2025-03-28 14:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945666-bug/1886543-stable/2025.1","cherry_pick_of_change":938827,"cherry_pick_of_patch_set":1,"meta_rev_id":"c592c50671dcab18b1ecbe458681fa0931ea5362","_number":945666,"virtual_id_number":945666,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 14:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-03-27 01:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-03-28 12:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-03-28 12:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-27 01:20:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2025-03-27 01:20:52.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-27 01:21:34.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-27 01:33:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 02:15:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-27 02:21:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 04:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-27 10:34:16.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-28 00:32:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-03-28 12:52:51.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-03-28 12:53:12.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"fbc945aff74bbaabcc94660c2d7dc8bac9835ea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 01:20:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9b80bded359fc91797bccf5a91c7d9f992513cd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 01:20:52.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1595/","accounts_in_message":[],"_revision_number":1},{"id":"d2e57e25bb485a0299957e232d6ecf704312cae2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 01:20:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945666/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945666/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"9066a1bd96d7b34b4e98f6f4d1a800fb4f7ff918","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-27 01:21:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ff05cd8c225172b2a66444273af785b49922b162","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 01:33:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8bc0b423f94c7c84e7226bf36175e1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0ed1e71861f4fb9ade952874315a751 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4d1199701d94484aa33f8fdb5de20f1 : SUCCESS in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d576a2065ce94c429367a3bc6adc1c7164fc2165","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 01:33:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7acfdb512be8c50d1196d91eb078938b96b8436b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-27 02:15:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945666 : FAILURE in 35m 10s","accounts_in_message":[],"_revision_number":1},{"id":"909eaa4f09f51e675dd209e2637ba0055ea30a1f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-27 02:21:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945666 : FAILURE in 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d78ffe29388fd98c4952317b3ba838d12eeeb486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 04:03:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d8dd2e6b4e0421f8a42d3a8c30ce4a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cead55ea634ba888d33ddde4f83ae4 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb4677f962644140884ba6f8eb3db2f2 : SUCCESS in 13m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/514e70f390d34219921a4440d5376ccc : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acface6f397146da9f3f3c4fe8139555 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/7f8588d3fa78491c86b92694a3a26ab7 : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15a2c46dd9d640ee8857f57fd2b8b6c1 : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8429c85240524b4f8175b263924102dd : SUCCESS in 1h 01m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b33e91ddd9bf4d508eea2aeacdb63c89 : SUCCESS in 13m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bf41175800724d6699c0c393e248a721 : TIMED_OUT in 40m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9888734af8984bed8cd25730ad4060a3 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b10402fc371b4b29b6ada9ff15ec7603 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bcbf004724d147baa0cf8f69bff11de7 : SUCCESS in 4m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2be26f5e1d364879920b56dd619fe748 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c18d70885fe42d4ae501961bf13106e : FAILURE in 22m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f6fa643101b4eeba8fde355045f2e9b : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c1cd9bdcf2e4a9d8b34e55a905b4fe5 : FAILURE in 1h 46m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8c5c092598d14ba89df7977fadc1947c : FAILURE in 2h 40m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bdc3b1336e947eaae02f8848ad52ffd : SUCCESS in 2h 21m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bcafbcf4534f4d8fbc9de23ecd1448f0 : SUCCESS in 1h 56m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b16fd3816a454f858e7e2823382e1424 : SUCCESS in 36m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c10a09b3d9d403b8bc6e6ff8cdb0168 : SUCCESS in 1h 10m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0be59ad16829427ca77bf0e7f4a12a77 : SUCCESS in 1h 38m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24f8a34d69d5446c8c844bc764c2f6a5 : SUCCESS in 33m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7fa7ac97b7747428e67ae00e6a48712 : SUCCESS in 1h 43m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9f28c0367bbd4d7299e98b5db8c2b543 : SUCCESS in 1h 44m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c8a0b930cbe4e079f2045a8f2df0e3e : SUCCESS in 1h 39m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/441ce6dfec5d437ba174722fae5f451a : SUCCESS in 1h 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10aa515f272144f1bf1727b4d2cf6260 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":1},{"id":"66d76ecc5fb269a06ad19960265f8db9c6689530","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 10:34:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"123c8e4640bde71a260d8187279d6a10d013ffa0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 11:09:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0312e3904ad60e902083a05306b5b5a713d1daa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d5103c85f14169a1b2a7088b06469c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c805e9748eb45e5a632fcea143eb132 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e72a0c6946d4252ae193ca1d53c5dd2 : SUCCESS in 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b104fb08ce9626f33e9bb5dda3e39c2825603add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:49:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49f2c5c535b2474b8d17a94614cec8fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5cc267683b94df291f7b2332797b331 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/234c466896b64c2ea85d3461e847867b : SUCCESS in 14m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59b7a960ce0d4a70ad84a868a3d136dc : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0511feb07742719b9c4e668bae2ae2 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/bf07c68488a14c1db14b8790c95e6f9b : SUCCESS in 30m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57429aa95d5249dc998edd02c958431b : SUCCESS in 52m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/519e941c64424b559e00b3e574539829 : SUCCESS in 1h 20m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ee15227ee9a460ab54a979d22aa8143 : SUCCESS in 52m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4087f42448a41c2bf4304eab8802260 : SUCCESS in 15m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/72f12923d3f041cdb4ab96a134376992 : SUCCESS in 22m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d0a53db56d0a45a5b52d5a4c2e390f12 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75889454a2a3461cb11802afc716991a : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48bf585e55374360a9b9fec2a1657342 : SUCCESS in 6m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/686dc145d72d4c1cbd6a999acd183f05 : SUCCESS in 5m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33286e83e29545e8a5aba0f976c074e9 : FAILURE in 33m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c918e0946ac24e1cbcaee86969f1288b : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47ccf77e818540179047f7d0773e8c28 : FAILURE in 1h 58m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09600a17b64344d1932141112c89f787 : FAILURE in 2h 34m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa18b10b73c04ed7a5417449b933942d : SUCCESS in 1h 55m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ae12b75ddc6648f2aa255cd506c45921 : SUCCESS in 2h 02m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2e8c7dcfe1b842f8ae85c8c3aac44984 : SUCCESS in 36m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c60252c3b4124a50be8bca9b1c042d0a : SUCCESS in 59m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b5d0530d9b734349addf09ae98d5029e : SUCCESS in 1h 39m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e949d8d4fe0c4407aa1a7c64fe6dba82 : SUCCESS in 39m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9d703633a9a43eb97d3174585fcdbbf : SUCCESS in 1h 39m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9ff9410a1b124446b1423aee606a722a : SUCCESS in 1h 33m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/24b8a470d4e144a18c6f6450ce5c3981 : SUCCESS in 1h 07m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32c6980d2d52454e820326e5a565ed2e : SUCCESS in 1h 36m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cff8d56a50fb4197955fad187e64c45a : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3547b807549b493b1b556c2f1bf747a3adafb51c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 13:58:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e62aa1b058847663bf46ff8af5c6fd9f89ff83e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:20:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d5543bde13f46cda96f5a473ed50954\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3e201e19baa4a6785a3ad428c21542b : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee4110a15b074fd3888a749fd97044d5 : SUCCESS in 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaf774122b4685450c4ee7e89a39f705c33e79e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 16:39:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85f829275b1845c3955e70344ff0b86c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7df36e613284d26b832ebb9ea44e562 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7169920743484c52a7d8e560c3259e82 : SUCCESS in 12m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c83fac3d292e4f1b9d9b8a53bc5c3641 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc03114c29343759d124e240870b333 : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/57a5d375be524c4b8063144f7ec63e81 : SUCCESS in 53m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0915fff9668e4a8ebee2cd0fcccd94c5 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2737463a276842c082a917e602a28638 : SUCCESS in 1h 38m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b973e304a6574c64a12e0a08c677b561 : SUCCESS in 1h 03m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1497af3d214f53a23d4adeea01b1ad : SUCCESS in 18m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c5442c7cf99462296a2c5e8d9d89204 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da61a4f41bb440178afd9f925e942793 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f988ccfe7e0475bbb931667bd0be9fc : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/968413f28f6f47b29b737f038e7d9f77 : SUCCESS in 8m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dd572e4b3e704b87a178e15bb1b76bce : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a3a67a07f65348a2838f2b64c8ae46d9 : FAILURE in 22m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d88bd84517994afd9927a8227c3bb445 : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/12631e66eb9f46a9a650d247f980457b : FAILURE in 2h 04m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65d954beb6a1486b972c6562c199d5ef : FAILURE in 2h 19m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/29e45d671b214019893ef71c6bde6966 : SUCCESS in 2h 07m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/da3114c616a94ee79a8fbc5ebfb7ee94 : FAILURE in 2h 33m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ed3f9902975a42299db55f9ace7f6d36 : SUCCESS in 31m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44523c92c7ac47c689cc5e2b0c27c121 : SUCCESS in 1h 09m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e73b645f418a482aaf8081da8aa818e0 : SUCCESS in 1h 23m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2b99a0b682704bbf884e518a75d0d24b : SUCCESS in 44m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/796f12f3d54e4701bc9649e2a0bd8d31 : POST_FAILURE in 2h 15m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/246ec097c4964f1e8bd63db6f3d0b11d : SUCCESS in 1h 51m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef90def565764b30ba666f282a2946cf : SUCCESS in 1h 52m 37s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61785a6b54bf45b49576587cf1043e76 : SUCCESS in 1h 44m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c1749de5478495bbe3364ec9f3454cf : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9043636a09c96d25df370984dc3609321d6af2f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-27 17:59:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65bdd2a648228afb3c34920ff36a50e2ed17f61c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 18:27:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6c56dfaff664b4dabbfc8e3e18bf665\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f4f8ddfa3864564a7b5edc9b1ad048d : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c2fdba969174a85a45adab9a376ca2e : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d6a06dbc3e2a8d4cace284097a79b4244da4533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 20:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4041ce86c93c423babe48b3861f078fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4f7f5a7141428bb27d061778065e39 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22bd82c735bc4c318de870462b75210a : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d709c2bc19c343329f6fd917bd5a5c72 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6510f58cd8024abead9c425db556dd11 : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/49dcbd9e0149470f9646d0717dc5e084 : SUCCESS in 58m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8cccf2de2a174b29b645bd0253e8de2d : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1bba3e620d064839bb07110f78e34165 : SUCCESS in 1h 30m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83329f1ec767426e877e5808c0fc7be2 : SUCCESS in 55m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cea12d0b30d41d1abd0cfd8c0f94d84 : SUCCESS in 16m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5542efa21b1e4ef38d0e7e3c39cddab9 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b72ce362e5a845d0b8b2ce5792637462 : SUCCESS in 7m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dea9f583d6748f396afa68aedce993a : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6f92aaec99c4802993e8d610ccc65a3 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/102dfaab4bc7477f9a1f575ba1f59f5d : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afd4833704f44af9b4d45dd54677e364 : FAILURE in 20m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/915f00a9bdb7458581efbdb602927bbf : SUCCESS in 9m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48cf8930933943dd9a7bb4009b3a46be : SUCCESS in 1h 54m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6cf042d34fdb42538d97cf2739ab2830 : SUCCESS in 2h 47m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/750efbd8a9de4def840d168e109083ee : SUCCESS in 1h 55m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/84885c2edb644416ba8d34ad781de55a : SUCCESS in 1h 55m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74bffe82232547d3b2b4f60be4e3adc5 : SUCCESS in 34m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5063778a1be041ab8dee01a830ba6ca2 : SUCCESS in 1h 06m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/799d29c3330c4f7fa17b507377bf2c80 : SUCCESS in 1h 57m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6a0f5050e5ac48608c1575dc7ec60baa : SUCCESS in 25m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8fcf0d297db043138f8a953362277bfa : SUCCESS in 47m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43978f4e6e8947c1abaf016c025c1568 : FAILURE in 1h 41m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/175c0325579c45e3877ed35ee606883e : SUCCESS in 1h 31m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c7ca0e7d33b44b8abea0b5388fe6430 : SUCCESS in 1h 32m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a6c653b5d7f4d2ab927874ad82bc197 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"206d885db1fec845927203cb23e2959f71ec0c75","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-28 00:32:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c017fa208bfba7b0ba9e3dc5e9930906aba37a97","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-28 12:52:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f02e13ba2fc404065fb8396f2828abf20426cd3","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-03-28 12:53:12.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50a91fd0110892fb9ad91e76c72028240105f0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 12:53:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cf567789a9f149ebe8a1cd384e07cea3bb2373c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:39:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afea03a75204600a458f1e1bd2ffb3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e1956129ad4494ebe0f90bd286704d6 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccd0a62fec304507a892aa8ba9e062a5 : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de263fdf210c48c59302f34f844d78a6 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afaf842c51c4e30a84d3d078c0bf958 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/bb851fe28ef742deab648bb626af0cf0 : SUCCESS in 41m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/049d1fc02e8648368440b51738fe48bf : SUCCESS in 52m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/edeb4c12c8134d42b3eccbd8ab964c79 : SUCCESS in 1h 44m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/600d4e18240f46ed86f773d5440bfefe : SUCCESS in 1h 01m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb91706ff0b7494e80e1b6dd3bd61618 : SUCCESS in 9m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4082f9197e604ba5a05ef7555c58aa00 : SUCCESS in 1h 06m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5f5db9c6e7c4866a429fd3a740c0afc : SUCCESS in 1h 17m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/505e1058ef3c4b94a73432008627dd46 : SUCCESS in 1h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb32e018ec444ed82e07d8a5250ccfd : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e148ac8bbac74b485f86c3deefd2692c051d33e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:39:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c592c50671dcab18b1ecbe458681fa0931ea5362","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:41:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7e3613f1fc4c7b8b2f06dd3f372180\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d94cec05e27f43f6b6125f724ae8f2b4 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb50012eb5174a7eb829371ae705ebe7 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4acf0a3b6ea89c98ad3f1052c7345b0b67a5aafa","revisions":{"4acf0a3b6ea89c98ad3f1052c7345b0b67a5aafa":{"kind":"REWORK","_number":1,"created":"2025-03-27 01:20:29.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/66/945666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/945666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/945666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/945666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/945666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/945666/1"}}},"commit":{"parents":[{"commit":"6d33ff2cad4a96ae6db67cf99018c20e18c37de5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d33ff2cad4a96ae6db67cf99018c20e18c37de5"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2025-01-09 17:45:27.000000000","tz":-480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-03-27 01:20:29.000000000","tz":0},"subject":"Driver assisted migration on retype when it\u0027s safe","message":"Driver assisted migration on retype when it\u0027s safe\n\nThis is a revision of I2532cfc9b98788a1a1e765f07d0c9f8c98bc77f6\nthat corrects the issue that forced it to be reverted by\nI893105cbd270300be9ec48b3127e66022f739314. The revised code avoids\nattempting driver assisted migration when the volume has attachments.\n\nWhen doing a retype of a volume that requires a migration, the manager\nonly uses driver assisted migration when the source and the destination\nbelong to the same backend (different pools).\n\nAs long as the volume has no attachments, driver assisted migration\nshould also be tried for other cases, just like when we do a normal\nmigration.\n\nOne case were this would be beneficial is when doing migrations from one\npool to another on the same storage system on single pool drivers (such\nas RBD/Ceph).\n\nThis patch checks what are the changes between the types to see if it is\nsafe to use driver assisted migration (from the perspective of keeping\nthe resulting volume consistent with the volume type) and when it is it\ntries to use it.\n\nIf driver assisted migration indicates that it couldn\u0027t move the volume,\nthen we go with the generic volume migration like we used to.\n\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\nCloses-Bug: #1886543\nChange-Id: I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea\n(cherry picked from commit b8610a01b61cf86db2f31e26eb0d4dab4a812ace)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4acf0a3b6ea89c98ad3f1052c7345b0b67a5aafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4acf0a3b6ea89c98ad3f1052c7345b0b67a5aafa"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
