)]}'
{"id":"openstack%2Fcinder~945785","triplet_id":"openstack%2Fcinder~stable%2F2024.2~I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e","project":"openstack/cinder","branch":"stable/2024.2","topic":"bug_2101859-lacp-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-10-22 14:25:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e","subject":"[Pure Storage] Fix issue with LACP ports not being identified","status":"MERGED","created":"2025-03-27 22:37:13.000000000","updated":"2025-10-22 14:27:52.000000000","submitted":"2025-10-22 14:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945785-bug_2101859-lacp-stable/2025.1-stable/2024.2","cherry_pick_of_change":945631,"cherry_pick_of_patch_set":3,"meta_rev_id":"56e7e5278b4ecb3ea631257065193b6c201f8d7a","_number":945785,"virtual_id_number":945785,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-22 14:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-22 12:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-09 14:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-10-22 12:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-27 22:37:34.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-27 22:38:27.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-27 22:54:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 23:45:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-03-27 23:56:17.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-03-28 00:04:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-28 00:21:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-03-28 00:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 14:11:34.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-04-16 18:08:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-10-22 12:28:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"abfa2b051764d9a583dc410fd1f53d3882e6378e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-27 22:37:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a324399346055266cda665477e62c1af4b5870","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 22:37:34.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1610/","accounts_in_message":[],"_revision_number":1},{"id":"0fa2edd59ff6719fce26e384acabaf4c3706b41f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 22:37:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945785/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/945785/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"de996d1eed7860c030ac3a8b701b502466a91e27","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-27 22:38:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"854d5543082224036003b91757578da06d779b13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 22:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3cad955dc954c75857c656078342f56\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b40f17f763574f00a8a2c93d247bb930 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3747fce60c10470d888fc364eb8e9b53 : SUCCESS in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f1d0179040b7b6a41d60ac7a830753d151f7425","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 23:45:18.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/513/.","accounts_in_message":[],"_revision_number":1},{"id":"1c82fc74233f8125b85d8eb5c27a0ef5bb8626ac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-27 23:56:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/945785 : FAILURE in 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9c0fd36f0f590e85b34de0677d3c3b31c88a6499","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-28 00:04:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/945785 : FAILURE in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"11b3d793493ad26d86c22a6899a4e5bb004dfe92","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-28 00:14:50.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/598/.","accounts_in_message":[],"_revision_number":1},{"id":"9be236a44646579fffaece1b19f9cd7098bc0dac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-28 00:21:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a660222b907343bf846e75ae1a836760\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/65502f6914f14d9e80ef9ed3f05f5355 : SUCCESS in 43m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/baeb2a0188d944d48a5a81dd7d5b72dd : SUCCESS in 48m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/04142f479562444dae48f3ff7dd22d58 : FAILURE in 1h 43m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/046eafbb1f4149ba9a807769bdfb5fef : SUCCESS in 42m 43s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"a18ad378511cc2929ea0edabed4c255b1ebd7351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 00:43:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc3765c218d44dd99a1625a75569f21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1770011d092e437eb4f4c8322ed8ba18 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4f78f372044ca4b8f3027ce4603ad2 : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56547772abc14718be196032cf6eaa62 : SUCCESS in 9m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ced8eee891934ab9998f4183c696bd4b : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/351d59c16142420e9fcf542be35105ab : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/061da11e6f154249b55aebe2f0adfb1e : SUCCESS in 52m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fadeb522e739492898d595557ea71afa : SUCCESS in 1h 18m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaa59aa70f2c40cd8f247ed1afa0b242 : SUCCESS in 56m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d833a2afe754414a9ad4eca2a44da937 : SUCCESS in 19m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/09e16d5fd44d439a9a8fc4f08e789849 : SUCCESS in 14m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47e4501303b04fe69a3b24ad0b324524 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/854125e3eb994bec945f0915d906b80b : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/535930c9dcd8491d844b5f20135eb7a1 : SUCCESS in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7af5e1bbc95d4b8f85944f3df0c54b91 : SUCCESS in 8m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c478f186572744d4a6a1d2cc5924a163 : FAILURE in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0c135f4c5eb46fabcf71ad0a75c801f : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0587764ac95242d9a92e73da32979dfa : FAILURE in 1h 53m 41s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3179d1401c33402cb80be534723e576a : SUCCESS in 1h 16m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ca16111fc905486e960cd30a1804aba9 : SUCCESS in 2h 03m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f10af9800bc74280b93eb49c3d8bc7b9 : SUCCESS in 28m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/747cf55fab7847a69b81ce5d0423c563 : SUCCESS in 1h 07m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44c028e836804c8c9911bb61af222d98 : SUCCESS in 1h 14m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e9062cf64594738bee7b243e54687c5 : SUCCESS in 38m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a303ddca92f490b84eb68d15ac9d4f1 : SUCCESS in 53m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/85c4634c26f7467395ee07afe78e570e : SUCCESS in 1h 38m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c1b957096741168e7a2961020ca2c4 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"72416d0bba6d6ccfa1a364c8b15b0ed73e860727","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-28 00:44:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"325fe305153ee59ade227a375dbce777e9bbe6cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 01:01:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdeeee09a3b4eb294f440943eced472\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c349c472a5f8476dbc3fbb7cd3e93654 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0adacb013a54fbaa0fa32cc34561e4e : SUCCESS in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d4f8e775c6e811a563b750f84e9572dcf56aad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 02:38:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf0e830aca034f22a63056a9794834d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5566004cf23f4dfcada21a5ab9f664f7 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76422188ad6544cb9ed4cd60f7e4ea00 : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd7effbcb3a94705997d992f7125d4de : SUCCESS in 8m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b44157c57b74c69aefe5d4d6b6d6775 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0404fc4df24794a42e5de0b0157e24 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/f67a661bc06d49f49dd01afeae41d49a : SUCCESS in 26m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18ce70e69c564568b2bb2ac407de208e : SUCCESS in 1h 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4091febcc044a539641242e62f97646 : SUCCESS in 28m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c90a29e0ed84abfbced6164b2a6ef8e : SUCCESS in 15m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2525bf94365a4ca7b7f1d1f9e5dce2bb : SUCCESS in 17m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bbdbeb1e1f7249be8c321e051fbe5ccf : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3fda984a8f84977b885991f39d9bc96 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8fc32def3e3c46439018514844af477d : SUCCESS in 6m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ce645fa22efc4326991ef0bf6024ae3c : SUCCESS in 4m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/68def5666e3a4919bcb33ef4312cf065 : FAILURE in 29m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01513c3fddb04de4af3e2a538e6c99c8 : SUCCESS in 2m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58f3c1e658aa42e3bfade25d2d2130a7 : FAILURE in 1h 41m 36s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/111c19ab7117498bb5772f5ac7352893 : SUCCESS in 1h 52m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2531af77aaf145ca83ae8b473693807c : SUCCESS in 1h 33m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e32ca2e2569d40468ffd186fc17420f8 : SUCCESS in 32m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ac1ff05bd2034fe898769d9410913e89 : SUCCESS in 1h 04m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3bb2e0b86faa4573883f1ce9adf8bfb5 : SUCCESS in 1h 27m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c5c66fa3f8674c1cab58073a72ac930f : SUCCESS in 33m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f289bf7768b84fc085d093e27511ae96 : SUCCESS in 43m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/10d72e8deb5142fea543a9c1696f15ee : SUCCESS in 1h 34m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21a4e9a1981d4465845cc78fc0a249e2 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f56db91f9b913ce4f77ecf77261e81889745a1f4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-28 04:13:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e6169b5da329e2e4000bb32db6a8571e17dbad0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 04:28:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a53f4df2da54c5794929b02db715596\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4369790db2de402db422fe264a10a044 : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/905c9d801ec244a0bfb62f7bad5d98e3 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6282c59c1b2a9a67e4646728b1da6e117ff9828a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 05:55:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03d85211fe5646958bd8c84897cfb8f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635256c32ce241a594bc4224898d4af9 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fc7207ee24547b6a03030113b3ba73e : SUCCESS in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b7382f60cb54869ae307853816f7e4f : SUCCESS in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81e0713c547b481ba822c41291282fa7 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c7cc56dad24123a7aa40566c141a33 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a31c43043faa4bc3927492289bc5a60b : SUCCESS in 56m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee3971acc1a1454d8fa43da68f7cf933 : SUCCESS in 1h 22m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a974a84918f6444aa29c0d68112cf6b5 : SUCCESS in 59m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d501014fb5624e73972c85f0f0139902 : SUCCESS in 16m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0924e3b462aa482887788b111b649a81 : SUCCESS in 16m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/364b15224aa9419c8ee229cfdd8be61b : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c79f3aafafb4d07b9298188ffbcca70 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a9f00abc49814c51b9d6c25903f257e2 : SUCCESS in 6m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2eedfd2b76e041f2822f678fc1ef6d17 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3925c976949440129d9715c831dc96da : FAILURE in 33m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85b2ac3fe9b54bc49057065d9b5584b0 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9b6b40866724698862ef60dbfcdf954 : FAILURE in 1h 06m 37s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e458a39fd93b4f09b2fd5b9dbd363c59 : SUCCESS in 1h 36m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a30edc4a71464a76a9280a4ab2630295 : SUCCESS in 1h 37m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9df92671fd15405f8a8881cd691b96b8 : SUCCESS in 15m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c05d86c2795f465e87c6a7a280d0e5b4 : SUCCESS in 52m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2a3bbd01c1c0446c84ea06515534bb78 : SUCCESS in 1h 23m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/74f5c2a893c44b51807b7674a16c2609 : SUCCESS in 35m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3eb6856f07d24f14bad9637906348f9f : SUCCESS in 43m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/75848388e80c40c0b38c61ed4948d878 : SUCCESS in 1h 31m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9c29d917497434bbfc23cfbc0e3e5ea : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8e76eafdcb1d854c2a8007e2144214dba4eef2c4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-28 13:34:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f6cb4f9277a4a140f540dabd577db81a468ccfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 13:47:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1df79d40257944238149e892bda91a31\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7119d0f693fa4769bb7877163372dc5b : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75f0e9127e514374bf5237763d8bcde4 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae89949a6e09fef392e987bc9f71de60819f7597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 15:30:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c44e42a8620b42e8bc2318109d7fd681\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d104dd369347b1b36cadd3270f32b0 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f56745d04ba14a07a7d8e9ae05694441 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a68b92e74eae422797a914295b5bf1c6 : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/006f39d8786e40d79d43b4ef134ec509 : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b202ef790d4c4d45af4874beb6938f6c : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/584086d0cf9641deb3e1ba0ff0aaf097 : SUCCESS in 53m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/704c94ea61954a3bbeda3a97c6df718b : SUCCESS in 1h 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cbef6e4f5bc4c69a9d7fbaf879508a6 : SUCCESS in 28m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a88d1193eb084569bbc353542b795dd7 : SUCCESS in 17m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7fbfbcef00d44758ac5ccda96f36cfd6 : SUCCESS in 16m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63acdaee112048b3ae9e5f54b3446c57 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef43d8cda351406e8caa05a52701da92 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a24167c520cc4518b7a7c58b6aa4c060 : SUCCESS in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/66440c7836774fb9b44b3a6fb0e32188 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf35060cadba4421a57621eb482f5dd1 : FAILURE in 31m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c88c783ca08d4f36bced2d3882a971a6 : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/502cf9fa9dcd4a9d889bb0c96715fdad : FAILURE in 1h 43m 54s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24ba2f416c764d3bb653cd375f61e4c1 : SUCCESS in 1h 44m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a2d4d789acb848b8b52a7876f6f0ee43 : SUCCESS in 1h 50m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5fdb6bea17cc4d09a41a582d58448369 : SUCCESS in 26m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b77c0b3d5ee14639ab565e10dc2d8888 : SUCCESS in 1h 10m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8e850ff8abe4790b3465c1c85de8e83 : SUCCESS in 1h 20m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8e9ebd639c6c404998a7f480cee32f74 : SUCCESS in 33m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/56b6d687d27242e2b682c2c0bff17174 : SUCCESS in 1h 32m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/be0282c937f64dcfbb4e20ecebd549e0 : SUCCESS in 1h 34m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1469993a19240b886fcebe6bcc27d77 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":1},{"id":"600ca9e99ce438a1fc01bef9a29f778fdc5ca2db","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-03-28 15:51:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15fa3398805b5655f38ea78b423c1852cf0c3127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:05:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a204e3622f40a1beb0910d07f1cf24\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e63d33a11154cadb2e9eb7cdae88648 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de12e39c188d41dfa88b9027129dbeaf : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c36b93f08f9b19ac6e20639f000ca0810bf77fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 17:48:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f05e08cf9c84a30be17cf118a83016f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f1467733bb49279c9ec98d5b3fabf9 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9a0175664774a75996d0f63a6b22401 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36a6135d8c134e7f82ac0521e0830b50 : SUCCESS in 9m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e341997bfe946e2a6c17c7d8d15dbe2 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196b56963ab4496d8f9ad44f32b6e3ee : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/da7e0ec7ab9549ea96f36edcbb9bd775 : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff71c30cfeb341d88cec3a3952a208d8 : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eba17098c0bb4500ad29dc20d4ef3c8a : SUCCESS in 1h 01m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c8ea4b761f8454ea340426192e8d2d8 : SUCCESS in 18m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d88e8474b9064dfb929aa60fb7f0dbc8 : SUCCESS in 14m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3bb25b60d88246a1a4049485d05af3c9 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31f5f9c738c6444786344b31ddd67ac2 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b5470ff7f164651a73946b6ebd2642b : SUCCESS in 4m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dfb81160ac204f589a09b00c622c8095 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f969609bc9a74b7793820ebd0fbfe213 : FAILURE in 34m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff1b10a0ca5d453090ef1d29abbf648d : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c7bf7695baeb41d6a2fe74defddfc1c1 : FAILURE in 1h 55m 49s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4b565c8cf864213891bf90e8de3ac14 : SUCCESS in 1h 29m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31252e67ccbe486ba7ddfd238b0d2342 : SUCCESS in 1h 38m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f0d44bfdaca84271b612770fab679ab7 : SUCCESS in 23m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a63fb04ca2f24bbf917b5665a1b2cfa7 : SUCCESS in 1h 10m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab04ba726b514a4c999095b75ee7cedb : SUCCESS in 1h 17m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d5107db23fa4e4d9610bc15d79952a7 : SUCCESS in 35m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b449c162935347e5bf8d112271c670a0 : SUCCESS in 1h 26m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/af2744c34c144b1b8066daea64914f51 : SUCCESS in 1h 40m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca845c2eb4a94c39826985d6cf6365df : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"550e867ab6d06dc1afef7903bcd234bb983d7da2","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2025-04-09 14:11:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17d20cd4f23311d31cd573f5feb5d41892b9906e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-04-16 14:07:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4df0281471b1d21edf6e27a42f301ff1eb4fa6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 14:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a2a358fd4ff46e282bc82262e812feb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4848b9baed014994bebdd80341f4fdae : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1c11666a7a74b6287e1089bbac28333 : SUCCESS in 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57e06b8b6bc03ccd7ba2c4148f522dca36449e2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-16 15:45:15.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4000c4587a2f45fca32c9f958d47a2f4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d1b287d65ca74beea8519ef6e4718dbe : FAILURE in 50m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6f58371fd364948a9bbe17000fe028a : SUCCESS in 38m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/744e55b7e6b242e498418eb35e76b31d : FAILURE in 1h 37m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07efdf2d7de44955a7e1bc5951a05b36 : SUCCESS in 44m 48s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/j... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"e8f73012246ed85a39607c76e12377f9b4f26d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 16:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da416739e1a341dd83ecb4211d2b902c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685cd473af7a45bba0b5bbdb5eb2821a : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a7ed16b99d64bdcb374b110e555c168 : SUCCESS in 11m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbb2cb635ba8499cb607f3133ac692bd : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9608dffa3c114c59aa2e0b9cbeb5bc8c : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb0850d3aab438abf76a779851123de : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/02b51fbc5b714e9c90fe7bb625e6f626 : SUCCESS in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fddd8e05c1f942f18f1038eee58af355 : SUCCESS in 1h 28m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d918d7bbb815496e8d890a97102e8fe6 : SUCCESS in 55m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bf23e87b3d4423c854124982b2687b0 : SUCCESS in 19m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2279882235c246309d17528e0d2a7a3a : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/280d085f14d641b7bab95f87090959b9 : SUCCESS in 3m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6224dfab9fce47daa982e71c06d7c437 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea741c6d83d549cd8463d6a2c12ff0fc : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/027f0792e3b749359b9b0f2911933beb : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b3b4de08abe54cc0945fd4a4a388a980 : FAILURE in 33m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5093c45a687c441f87bdced43e2cd005 : SUCCESS in 2m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e44b4241b27745fba48187088f5cf849 : SUCCESS in 1h 50m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/95b2967b5e484fd1a3b68d927fead9d1 : SUCCESS in 1h 32m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/48d29ecb52d24c28ab9b3534ff2c7de4 : SUCCESS in 1h 52m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1b21e5c0fd9449b7973db32ca241d44b : SUCCESS in 16m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/695b72088dbe41c58ea33b566e991551 : SUCCESS in 1h 05m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d79c9ebaeeb42129c5c0bdcd2e793f5 : SUCCESS in 1h 32m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4507dc40d0af4384a0ddbf98119a762a : SUCCESS in 36m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ea80c2a002104c8ea0b6f1bd9ac0b32e : SUCCESS in 1h 21m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9ad0ac83b3b64b15a26658fe126060e4 : SUCCESS in 1h 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73e436b0c6c040f0864411b7ba176e13 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4aae4d6dff78eba8047b29963a99b7849c951651","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-16 18:08:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"16651615cf205713b0605258c2b7c3349f631084","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 12:28:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba228dfc323cd9cc83d2a762915c49c93c58f26a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 12:29:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a61d8ed76cd3c97b63e0998a3d62306129b20738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf1850a75be04793ae41b766395c53ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e3b9e29eb941eab30e9be90b8eae16 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46fa54a892314b508735825794c6fba6 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/653809879b6142c394d2d04facb5ca3b : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61946efaaff442258fa5f74a6aa5cb15 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/148283eca01e438488873a84b28aa4ee : SUCCESS in 1h 10m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f259b57a5674782a3abc59698654d05 : SUCCESS in 1h 29m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03b61507ee044ed3b4e5974e13d5885f : SUCCESS in 1h 00m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978451d31f174cc780068e7411514ef8 : SUCCESS in 15m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6732ff52a5c44519eb51c6c8a21ab7c : SUCCESS in 1h 08m 46s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9873e473b8484d97aa29eba65a3f83fb : SUCCESS in 1h 07m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f48c462cc88498bb0d43fd89e937b42 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ea2b4a5c1d1b735ef9b6353da92541cf7bb57e95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:25:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56e7e5278b4ecb3ea631257065193b6c201f8d7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:27:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffc38a5a30d4469c823dcc03992c4234\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64a8f0664ae435e9eb6a036a05cf0a6 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0854cb89c5e043bea2b30215c05a1879 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9ac7315e6e7b46534bef72918a8e972270e012e","revisions":{"b9ac7315e6e7b46534bef72918a8e972270e012e":{"kind":"REWORK","_number":1,"created":"2025-03-27 22:37:13.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/85/945785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/945785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/945785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/945785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/945785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/945785/1"}}},"commit":{"parents":[{"commit":"0b165a3f3ce49a25d6257f266aac702e28213590","subject":"Dell PowerMax: Add PowerMax Laurel Release support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b165a3f3ce49a25d6257f266aac702e28213590"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-10 17:31:56.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-03-27 22:37:13.000000000","tz":0},"subject":"[Pure Storage] Fix issue with LACP ports not being identified","message":"[Pure Storage] Fix issue with LACP ports not being identified\n\nWhen using iSCSI or NVMe as dataplanes target ports are identified\ndirectly from the backend.\n\nWhen an LACP bond is created this was not being correctly identified\nas valid.\n\nThis patch resolves this issue.\n\nCloses-Bug: #2101859\nChange-Id: I5b56c590a6c9b82ab2e08e4211bfd3d187afdf8e\n(cherry picked from commit 5070eeaecf2837c8f5109c63518d87a08f9c40b4)\n(cherry picked from commit d5da037e6c012312d12d224df5a19385c29d7c66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9ac7315e6e7b46534bef72918a8e972270e012e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9ac7315e6e7b46534bef72918a8e972270e012e"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
