)]}'
{"id":"openstack%2Fcinder~946788","triplet_id":"openstack%2Fcinder~stable%2F2024.1~Ic3bc26eee419ecae7377f05e9488b809cd3b843a","project":"openstack/cinder","branch":"stable/2024.1","topic":"clone_qos_2024.1","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-04-24 00:04:19.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2025-04-24 03:31:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3bc26eee419ecae7377f05e9488b809cd3b843a","subject":"[Pure Storage] Fix QoS setting for cloned volumes","status":"MERGED","created":"2025-04-09 16:14:54.000000000","updated":"2025-04-24 21:10:25.000000000","submitted":"2025-04-24 03:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"946788-clone_qos_2024.1","meta_rev_id":"ee886878807ca1992abce3d747e5ba596505c9b4","_number":946788,"virtual_id_number":946788,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-24 03:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-04-11 13:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-04-23 19:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-04-23 19:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 16:15:05.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-04-09 16:15:43.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-09 16:16:20.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-04-09 16:27:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-09 16:32:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-09 16:50:44.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-09 16:51:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-09 16:55:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-04-09 18:37:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 21:38:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-11 12:55:17.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-04-11 13:05:29.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-23 19:05:54.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-04-23 20:38:34.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"b554da63c56f7d7df929e015f2577b04997899a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-04-09 16:14:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f7443f17fac70c930a11f8a879bf6780d904fe8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-09 16:15:05.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1670/","accounts_in_message":[],"_revision_number":1},{"id":"807b71d1a4e5b69e133b378c193bbb218a24de4e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-09 16:15:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946788/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946788/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"0dd9147c822d5368ae5d3e52861d72bfb524e7ba","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-09 16:15:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"64b39b3586ab73c866919644942e76571fe0a400","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 16:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/254423210a8b41fca5a73ccb3e7edd60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c416817216fb464c96fd62e5f6c91aed : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5db19ba6078c432990e8e6018ee4a071 : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8204b3bbcbda099d1d474d9169114dc9cdb691fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-09 16:32:38.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ba4eac7cae2149e29ffbafebf6953e32\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5570dc9295034cbcbb615b7df018acb0 : FAILURE in 12m 32s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f204c17d5f5242f8b30b7d7495882967 : FAILURE in 11m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c946b064e1044f5814c75f701fb088e : FAILURE in 13m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9574fdc68124458591d290efe453699f : FAILURE in 12m 30s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"525b315eb24f66325715ae0b2f4ed8e9fc0acb81","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-09 16:50:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946788 : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":1},{"id":"24cc0124c4ecc69b1e36378f221ad5a321cfff13","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-09 16:51:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946788 : FAILURE in 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b270fb450cc0215fa3b3ff621236b5e609f3ecac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-09 16:55:34.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi_DEVOPS_TESTING/5/.","accounts_in_message":[],"_revision_number":1},{"id":"38498a1ac988bd11b58cda410bd5848e2143d9e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 18:37:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f100385972bc4068ab824d8ea9dce4f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864b79a44c424ad1897ed60635ff9bcc : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d97f5b86299452897865868ca22113e : SUCCESS in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6738caa7cc244ee9463bd846d0482e8 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e6e9d9aecd4b05aa62935e5e19a29d : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/00a0a7b22cb245408abbe74091a51b56 : SUCCESS in 26m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/12d7452b9aca4237a52d4250a0146d1d : SUCCESS in 46m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b944afb88abd4e128271d3222976b2b3 : SUCCESS in 1h 19m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff80eaaab5254b2c9e031f3f4ff3a34a : SUCCESS in 25m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/933ade5bc3354dc9996b64c80301718d : SUCCESS in 12m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/434af3947fbb4c738ba0844d76a537cf : SUCCESS in 17m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9eb87da2c4a45f0b16201f73ae5d075 : SUCCESS in 2m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac544aee412c4d02a9c93455ec87d815 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec14c4e0a5e14358a127fd818ed96c7c : SUCCESS in 5m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d6adcc436f04c14b54f846ff2c963ba : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e7db499f1c9943e19a9ea0bfea8f8733 : SUCCESS in 35m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/faab226a3a654c449ba63610df16ae9a : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e01040068ad400d98522815ed7d9fce : SUCCESS in 1h 51m 10s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6709e71b2cea475d99835a282486bd15 : SUCCESS in 1h 33m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c694d0bb87714f67b9b8f0c9ec0b78d1 : SUCCESS in 2h 11m 19s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0d8ee352199e4c0d8d226ba019ff169f : SUCCESS in 1h 07m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af0f8eae58da4feab61fd007459c0623 : FAILURE in 1h 36m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/08cd08235dbf4624837c6ea33eb07989 : SUCCESS in 45m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/805ce8a322f54e548067859b40cb0787 : SUCCESS in 1h 04m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/19dcc8383edd4496b44aafedb9f6cdcd : FAILURE in 1h 13m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be763ce8de64b8eb5acc23a5b56c993 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1768e474da1211f0d2b13e348604cff2d1b82cc4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-09 21:38:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3214c8894a1348d62bc087fd7288138fd9d1436c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-11 12:55:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f6437907627c8f3a1d84338e592cb1acd854c82a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-11 13:05:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"598155aa1a52ae1c2cc972cc8536288d03c1cd6a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-04-23 19:05:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b2e5e161373d4d433bef0e240c92d4317fd9972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:06:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cda4bb575a78204573de114b84aac81afb471ce1","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-23 20:38:34.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bc756da0875ccdfab6f7bd5ec66587fce15b0bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:22:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a731e6b595544dab97524ebfa7db802b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8f95a68b954de2a2a4aeb73fc36cf6 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1b14b7bb1343f1a7dff2222d330e0b : SUCCESS in 9m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e1d8eba1d6f4ae488efe4ce1716d22c : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59d81273c9f47829747125596f86547 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/8c122bde553c46179e1c6d7f25b856ae : SUCCESS in 58m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/673a016fb83e41ecb8fba89365decb22 : SUCCESS in 38m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/793a6141e01e4079a60bdbaf7f0fa1b8 : SUCCESS in 1h 16m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f15076687394552a0139619fe213c67 : SUCCESS in 45m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ee107dfcf57446c951654dc7ceeb379 : SUCCESS in 10m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b7256dfe5428427d92acc0fb1f18ed0a : SUCCESS in 1h 12m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbc9d4652af74956a227265350d6a034 : SUCCESS in 52m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d25c824deca24e9382d374dceab86521 : FAILURE in 42m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a28c0bb281eba7c39cdd669329b00980cfe0b488","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-04-23 21:24:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cb28564f4b6e90f0f5a654aab72fed9c01ac1e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 21:42:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e4cfadc84247ab8e59ee9ab1674653\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecea414387d7494dbbbefb6872d002a9 : SUCCESS in 16m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c76fb477d0bd4e0fb2a79ea73335a1d3 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f386f6445793e1814f6a147b39120f3ce0697522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 23:10:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9f6b6e0d10c46a5b47c2a0423daf4b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a705590a5dc4026af884a9a915eafb3 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b39ce21aef0b40eead55af45690aed57 : SUCCESS in 10m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4203a3fd1f16404ca201aa757b07d85b : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9117a30d4e4e70ae7f5917dc9cc50c : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/7c40c08a90d541a4bf702950811f3d1c : SUCCESS in 1h 15m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0cf9a7702cd7410f933ae87effebf231 : SUCCESS in 50m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e3331def9104b3f9b23dd93ced0b4fd : SUCCESS in 1h 20m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ee8b518e64d4145833ba2ded390bbe6 : SUCCESS in 55m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb225216941a4c358becb8671693343f : SUCCESS in 30m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d18ec47dfaa4403090a4eb16699169bb : SUCCESS in 16m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8e19974c82d4b11a61973ca814c5739 : SUCCESS in 4m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7fd7cf133d98494781fe02d339e9098c : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc962a17e41f437eaf972458b7167e80 : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09dcea06e4e445e29e0c947372ef9458 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e9716760d3d49be909f24cf5a418903 : SUCCESS in 42m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/415841444caa4408869ada0c1400f9c8 : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/badfcdd93bcb4047bb3aaeaeeb7d0810 : FAILURE in 1h 40m 12s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/516dfa916fd140b69281845c04e37279 : SUCCESS in 1h 30m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a1e18b6fc34b478aaa3f3f83c80f996b : SUCCESS in 1h 36m 20s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc62d8d29a6844aca4bb9261a9dec842 : SUCCESS in 1h 04m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fafb5919fffe488795c85f972a474981 : SUCCESS in 1h 27m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f95b5167277a44518e8961c659b4e0d1 : SUCCESS in 27m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5606018303be44358283d9b3edf7d54d : SUCCESS in 57m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a624c67b292c444a8b509750bfc4ef0a : SUCCESS in 1h 04m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b2e32558de4445da1d1281f8bfa05e6 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f39b9293e31b43673bdecb33bfe10f2391d7897a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-04-24 00:04:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79b3bf86cca4827dbb51594435214a9eaddc74ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 00:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee0ff1e80a8843e69f424389d58c18fc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b28c9e09882b4d8094165d00eb8afa50 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bdad4fa4917243268682802bf759b529 : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a3ac9f467c83709123946750cd06e2de4680f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 01:40:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5ed6a8059746c491dfa8ca7ca38898\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335da33390444a458288472867efe237 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d24bbc9ebfc4c079b0eaab44adc8519 : SUCCESS in 9m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ffefc8c6f744477875061224aef68a3 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20a90c2d5e0413dad2418a9c7c77778 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ffb468831bb04fc491f7a5aefe81a27d : SUCCESS in 39m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eedf53cdb2e2408a9c0beb3d94e29602 : SUCCESS in 38m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aada6cc05b634aa1ab9611788bcf4f29 : SUCCESS in 57m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b79ba78edd444baa9aa85e0f55ce973d : SUCCESS in 39m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ebfb0dee318431bb47435c643c98d25 : SUCCESS in 14m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/79b40a4b286b4e21ae768854bf5627fa : SUCCESS in 12m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9238bec83fd34da092a22145be539e3e : SUCCESS in 3m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd83f18e197f4b53a6394d7014ba63bb : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6005bfd4bafd4944816aeac733f8b76a : SUCCESS in 5m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d00a4b47a04a4f3dbf23fa81c35f33f4 : SUCCESS in 5m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/31a75719dbe34baa8f6c1426d8652ae5 : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66ebfd7449874332a2e543a1ffa88b44 : SUCCESS in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d13857ae64de41d799897a28c67ab809 : SUCCESS in 1h 14m 42s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62c50280e9714774880595cd68c177d9 : SUCCESS in 1h 29m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ac5cf338455447e8a188f7bde9db9359 : SUCCESS in 1h 23m 19s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc17af5026db46b899df285c3214f03a : SUCCESS in 1h 03m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b36ac656ca849e8925fde3cda68903b : SUCCESS in 1h 15m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/77aea1449e3342c8b9b6a74e8c2d304a : SUCCESS in 27m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a7f32b5cb7b41f9b57f11b95948c462 : SUCCESS in 1h 15m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8d02fc24daad4e3ebf24f3f8ee7dcbe1 : SUCCESS in 41m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4508c14b5dd7458dac3d42e70c9b43d8 : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a2714a805bf9756e81559a9d15fddf61900ebce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 01:41:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f076399bc03ba16d77e637f9491119a36be4a7e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 03:31:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb611424fbe4d5ba4f8a8d1a3fe666c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8237209cde4f51a0107303ca4ba94d : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/935b5924ec7141d09e439c7ec9da41d3 : SUCCESS in 8m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/192a46a114be40ed993869a7d5f7ba25 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28531f77a322471cb1559741e82b98c8 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f86c38e45fca4b41b1d70ea0b3af5d2e : SUCCESS in 38m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f2656178ed24d138710b9f7726ee876 : SUCCESS in 1h 07m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58c79199e8e744a381460be903706447 : SUCCESS in 59m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b92980debef4567abdb551241787ce2 : SUCCESS in 40m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43eee109ce0b48d185d160cd555154c8 : SUCCESS in 14m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c78f60c99f1d468785afc221020fef66 : SUCCESS in 47m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e312fb87bcd84b79ac2a7a233f6bb0e4 : SUCCESS in 1h 45m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8351760900ca42439fe5d15ac44dc29e : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"88f4e0f60887084575235b3efd10a5c274b693ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 03:31:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f05b65556b55af2a9772f3878b05e75646d7cd69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 03:32:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a757e4d95874448abc43de7c189fadde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464965f696d54846b0a921c0038516ea : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/666e7ce945a849e5b8ecdb311fc7547d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"94f1daef58ccebb48500142f3df8e129560f9028","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-24 04:42:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee886878807ca1992abce3d747e5ba596505c9b4","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-24 21:10:25.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/3dc823ae97d24a69bea62dd5a936c1bf\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/7cd6af7d010d4599b63ba4281e672bd6 : FAILURE in 4h 45m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b571e5f54216820d74cfd0609a3b2b625bef027","revisions":{"7b571e5f54216820d74cfd0609a3b2b625bef027":{"kind":"REWORK","_number":1,"created":"2025-04-09 16:14:54.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/88/946788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/946788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/946788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/946788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/946788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/946788/1"}}},"commit":{"parents":[{"commit":"a54442aef0860876db85b29f8df1b2c1245aecd9","subject":"NFS update volume attachment format during volume snapshot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a54442aef0860876db85b29f8df1b2c1245aecd9"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-04-09 14:49:21.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-04-09 14:49:21.000000000","tz":-240},"subject":"[Pure Storage] Fix QoS setting for cloned volumes","message":"[Pure Storage] Fix QoS setting for cloned volumes\n\nVolumes created from glance images on the source FlashArray using\nvolume cloning are not being assigned the correct QoS specifications\nassociated with the requested volume type.\n\nThis patch resolves that issue.\n\nThis patch is standalone for 2024.1 as the codebase changed\nfor 2024.2 and this bug is addressed in a seperate patch for\nthat release.\n\nPartial-Bug: #2100547\nChange-Id: Ic3bc26eee419ecae7377f05e9488b809cd3b843a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b571e5f54216820d74cfd0609a3b2b625bef027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b571e5f54216820d74cfd0609a3b2b625bef027"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
