)]}'
{"id":"openstack%2Fcinder~946858","triplet_id":"openstack%2Fcinder~master~I623ca7f9061b3e3edb9660da65bd9d64309f52d4","project":"openstack/cinder","branch":"master","topic":"bug/2106680","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-11-12 19:36:16.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-11-12 17:35:30.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I623ca7f9061b3e3edb9660da65bd9d64309f52d4","subject":"Handle volume_clear_* opts consistently","status":"MERGED","created":"2025-04-10 01:30:53.000000000","updated":"2025-11-13 06:54:43.000000000","submitted":"2025-11-12 19:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946858-bug/2106680","meta_rev_id":"5b7bcd00c699879d1f79dc184cbdeb3d0f208088","_number":946858,"virtual_id_number":946858,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-12 19:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 16:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2025-11-12 17:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2025-11-12 17:35:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-10 01:31:01.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-04-10 01:31:46.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-10 01:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-10 02:18:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-10 02:32:39.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-10 02:47:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-10 03:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 03:51:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-10 05:20:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-04-10 06:48:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-10 08:44:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-11 04:11:45.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-11 12:23:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-14 05:08:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-10-31 16:04:42.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-11-13 06:54:43.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"ae1903f861b529dad2102e31daa28af52e2bd761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-10 01:30:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1fd9d2cf4737be635c2b4c1931e0e6193405ef4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-10 01:31:01.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1671/","accounts_in_message":[],"_revision_number":1},{"id":"1a1354396d75b0048aa4ee6e0e26abf24f7e647a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-10 01:31:03.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946858/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946858/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"c140cd3076e473297063099aeebb56a01c7d13ae","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-10 01:31:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8eccb20cdd0dbadb77d8ef506a00998051acf95a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 01:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d00018f149154593a610a6df498ad369\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f055032c72f4135b8557db3631a4759 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0de1f7e306c41d990c92a912bf1141f : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/932c3f4308dd475ca6948d01b0ddb4a6 : SUCCESS in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2beae0a8f530cdae5e4751b0ffd8907e921cd2e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-10 02:18:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946858 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"46e68fd2a2be51a3551904159badf14b92bacaea","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-10 02:32:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946858 : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9713a8674abf7d85e1457c924c4fc4c313ce9844","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-10 02:47:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee99c301caf943dfbe6ed182a9163b55 : SUCCESS in 1h 14m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/26bb1da25d0a4d20960bf3ac2bd87219 : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"89e51f796a3cca7be191e49010109b91e6467e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 03:46:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9acc750979e4c74ac47ad9733d1df25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481c5b2d067a44aaad6bb39d9de2f4d3 : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1558655f64b6475c85cc08c4233f788b : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce33ef0c96834c6585d5eae750e5b0da : SUCCESS in 13m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92b4123799ec4ed4a87b1749ee9f8684 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56130a2e5d144c8ab6497c55ed9696f7 : SUCCESS in 15m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/62fa764359da4e7e8300f8f927841ac6 : SUCCESS in 26m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93ff9ba3721045fda363ee9545ffc517 : SUCCESS in 34m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28430e1f9d1d439990711a99f714ec4a : SUCCESS in 1h 46m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ab83f69c88e48cea71ac540c6b827c3 : SUCCESS in 1h 01m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3bbdf351b4f348959758fbcf023c063f : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/416db2f115f748959faef441ff6113b3 : SUCCESS in 4m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23ceaf4259f1432c8615fcc6b12cc304 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1a3d9ba90b7490cb4b7807127eadaf0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a85371bf5a4b46e99be0a9de8138429c : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c45525c0a86408d9886ecd5b9fa0caf : FAILURE in 26m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e2274c083be43da88373f27c476a92a : SUCCESS in 3m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f90e672f123e4479a39a5376f615fc03 : SUCCESS in 1h 09m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/41048d0386b24728a79fed1fa03c7de2 : FAILURE in 2h 14m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2923713d85f64bcf91448a6e00a04186 : SUCCESS in 1h 38m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1e59ddb9a1a64a73ae273e706fb9efb0 : SUCCESS in 1h 57m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/edf359f6b2324434803828d0e6deb246 : SUCCESS in 21m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6381649308a34976bbf04214403d1d5b : SUCCESS in 38m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8eb93f09b474662b33b3864158a94fb : SUCCESS in 1h 35m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8e521e0ef4347b7a33969b06c1d2ac9 : SUCCESS in 37m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f62f098d32db41feb9a78e8d081fcd80 : SUCCESS in 1h 29m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c4eb8717423e4fb0ab111bbeff318c04 : SUCCESS in 1h 36m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06e80c4155824de4af153424f2657080 : SUCCESS in 1h 52m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b2cb10b58944ee58e96ad77c658a7ec : SUCCESS in 1h 22m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a962cea76b754775a455ab0f0238b961 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f7b037ff4ab740b0c11516d28ddaf6914de1da21","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-10 03:51:03.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e4184d2798984075aba22c331aa0caab\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19b40b97671c4b33ba14e1ca2705b2dc : FAILURE in 46m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc1d83d680564ef99b942a7c70c63eee : SUCCESS in 39m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c1bd28264642486aba86f0a094445763 : FAILURE in 2h 19m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/29c2131fdb084bea83ec7cf6c90e2e74 : SUCCESS in 46m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75202d4b6a7f4204a24a53e8ec5acbe3 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3300e82a76264ed481c72f0a54695832 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62a3b3ada3104eb5b3404cc5eb95e698 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9c36ac5f944408e9a5eb8808dd5d1a4 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-d... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"fc3504a083ae088559b0c1702b45d877c3371f7c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-10 05:20:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/1/20250410103055/NEC-V-FC : SUCCESS in 3h 48m 26s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/1/20250410103055/NEC-V-ISCSI : SUCCESS in 3h 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d4fa1fb730941750179dbea8160afb522250dffb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-10 06:48:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b51b47a2289345d78efbb29871f08557bebd3b29","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-10 08:44:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/1/20250410115214/NEC-ISCSI : SUCCESS in 5h 30m 26s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/1/20250410115214/NEC-FC : SUCCESS in 5h 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"478f16d861d0da9c36266d538784a5b3ea773eaa","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-11 04:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/946858-1/HPMSA-iSCSI/ : SUCCESS in 1:58:16\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/946858-1/HPMSA-FC/ : SUCCESS in 2:07:10","accounts_in_message":[],"_revision_number":1},{"id":"eba27b98bf34270518a0498e906c2a024f821a7b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-11 12:23:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"413f73e13ba5f4cb8928a9ddb69714685dfcc9b5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-14 05:08:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72e281321a1a744b4873472464fed319e11c6119","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-14 05:10:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cf0ed000dbe5b7970cc04ef257ead5f8e802199","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-17 20:01:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c43d8492d668bcc8bcb2b774c37e956a1b86cab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 20:02:03.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1724/","accounts_in_message":[],"_revision_number":2},{"id":"70fe0e2f2d28a559b2856a9105b88746f3e056c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 20:02:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946858/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946858/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"12d60547f9a2a907c724ada9395ae6abd23430d7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-17 20:02:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c3e17e735509d94d00fa4822bb3dd4ec4981441d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-17 20:48:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946858 : SUCCESS in 45m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f58bf6ed539313bd3ce752b8eca3aa376bd0522c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-17 21:03:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946858 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0e6946eca3be6e44417b2788fd989aec41eb2451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 21:04:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d047da8aedf40cc8ae96ec88e9c83c9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ed7f7bb5a484365ae6deaaa33303a10 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53d6951ba5db4dd19b639092a92fd0eb : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/75a9f469b24a4f3eaf8731c80f46b5ae : TIMED_OUT in 1h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f9d592b69909b366cb5fc124b78890ea3582f15","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-17 21:25:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0e1cf15ccd4a4b7c8b59ac115ae36fd8 : SUCCESS in 1h 10m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/afbacd887a484a7d9dd3b780d44f6a5c : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":2},{"id":"bd4150d9e60015bed78000ca3aac755263d37111","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-17 21:42:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/527ad1ccb9964c5da45c2557219f04ee\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74ab186c123b4e3fbd9e8008856a088d : SUCCESS in 45m 46s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4041ab3a32e54f999bc7735bc51d6bd7 : SUCCESS in 41m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e187ac560800425592a80ff454b9975d : FAILURE in 1h 40m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7ba57e19ad1349e7ae05a6cce0a0e0f4 : SUCCESS in 41m 52s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fb9391f4f7d4a0c93849b73087bbdc3 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f91b41785804d83a230d401d08ade0c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c305259776514c1ba07c29beafb30481 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/438d770de35f41ee9aadf471e2374bc3 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"2d055e847466cba2cd38d96b7b1d4fc5b6ae12b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 22:21:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/282c5b894a154ffb94380128067bd77f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fefe413bdf4718968ea91666b5ca4d : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b09ecebc1c842f1af627a17fc54195b : SUCCESS in 6m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c422d19db3c1486fb39a96ecbee71b91 : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4a36d24faf04ab38b58e1f737e59658 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b0ae7adede4958a705761d7016877c : SUCCESS in 15m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c4587565589b4f42b9202fc42ea9c57e : SUCCESS in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f5bcb62aa81e48f3b717125529c04f00 : SUCCESS in 54m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2aa35167c95c48c4b5d185fdd539352e : SUCCESS in 1h 48m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8657c6b13b84a22aeb7c42b8da14f5d : SUCCESS in 36m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bce1c4be583f434aab81ca4108dcc97c : SUCCESS in 20m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df00d5fc96ee4b2e891b84e6038e4504 : SUCCESS in 21m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e36c2d50b9234ac9a4ed6861ffd38df1 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ded0622c16e644a49ea8315fb06dc8d5 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9323837359a4105911476fba566dbfd : SUCCESS in 7m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6e51eb1f59a94b568274442b81201efa : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0670e950587f4b73a35bfa8b7be746ef : FAILURE in 26m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/436e2ff25ec24f368ad2f2654ea4d25c : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/919cbc8a1f8842d7bd4fcd2c7a6bb830 : SUCCESS in 1h 46m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1986c2b8ba2445108e10519327b966e3 : FAILURE in 1h 52m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89617b8af86046d3b747b1eb367c1470 : SUCCESS in 1h 46m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/52ed910bd1d8479baea10634740995f9 : SUCCESS in 1h 31m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0ad090c37dcb4b34b21729edd3dda942 : SUCCESS in 31m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/084c4a238ff04c62a4f5bd621b2c0c43 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bb38b38ad69471d9546b26d6257bbf3 : SUCCESS in 1h 29m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8e78590bd1834d84a38a6c959c2815be : SUCCESS in 22m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d6323705ae2a4640ad8b2fa83f347090 : POST_FAILURE in 2h 14m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1d6a280a29b8401e9a3d049d3ab52201 : SUCCESS in 1h 42m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/926a8641656047b990e2abca5669e986 : SUCCESS in 50m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de0d63852d614a6caf364f85e0bdeec1 : SUCCESS in 1h 22m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43e988f21cf54d978dd9b81301fb327f : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":2},{"id":"631bb7cc6ffe8cd100f2df9ef6681db50b0a3580","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-18 10:49:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/2/20250418153239/NEC-V-ISCSI : SUCCESS in 4h 06m 20s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/2/20250418153239/NEC-V-FC : SUCCESS in 4h 16m 49s","accounts_in_message":[],"_revision_number":2},{"id":"157ccdd3a8f675ee1c6be493618765449dd330c8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-19 10:04:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/2/20250419130927/NEC-ISCSI : SUCCESS in 5h 30m 11s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/58/946858/2/20250419130927/NEC-FC : SUCCESS in 5h 54m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9cdc30f6e0f88aeb949397435c6fb68151e911da","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 00:18:02.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a574fe7431737311d43fc26c525209fa6de049f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 12:46:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aca3c6d55f29e092a6ce5c1fd72cbf22bc0075f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 13:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9211ecedbcd3485b8530faa23f93f3b7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/887eb82d5cda4a379af9a318b028eb63 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f0244719ed4480e8e4aedb54c579ca2 : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/79b6af25641445a4b42d71b0d5f19d5a : SUCCESS in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"66ab15363f6bbdc9ebe00817aa286ef6898f1583","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 13:34:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8335443c8025f803d0b73fae5d338ad71574a8b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-25 14:07:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d335271e80c74c459a3747958d9cc158 : SUCCESS in 1h 16m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7b9c6c6b2be04bf081fcd1ffaea03a4d : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"008740be3a732a3bf4e7426eb1def5c3429f2b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:48:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c640a8a655944dc9bcfa7aa2d9f37cee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d168bff9564733b8c82aeb4d7fd948 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/643b268a438842688c7c4c8d2b5ae712 : SUCCESS in 7m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4594422f5bb4170afd8da83e92bb7b7 : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9fb8186afe64c00a220f7d2e52f7584 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7c70def7b64f15b6fb3a9ba90684fe : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/f50df15aae9142a683c791675a86fb1e : SUCCESS in 43m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aacf7109a3eb442f92fc2b933c49cf2e : SUCCESS in 38m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc0c313b8df54436924a20f33fd6032a : SUCCESS in 1h 31m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c6863f4a7c3452aade5d4ff72565d0b : SUCCESS in 49m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff636a18fb04e0f925bfb85d5164254 : SUCCESS in 14m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7adb9984c34449039cbe11a0c51f5edc : SUCCESS in 16m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87a6e7eb992f44af9326a386e6e9bf8f : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6c3d17bd2374ba3bc2c73b7f62b6980 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac036fff516449bc9c424cc0c51eb4cc : SUCCESS in 3m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d62ea986eeec4672a1a01de02178d702 : SUCCESS in 3m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/229c371cee5545be88b308f621f38867 : FAILURE in 34m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81300b494baa47babc7b5169698b481c : SUCCESS in 3m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2cd6a4222d0947d1b7a8815a1a21ce78 : FAILURE in 1h 35m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48957f621a9b4bfb87673789c1fcaa3a : FAILURE in 1h 31m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7043a578ab0c4954b6eebe803f03fccf : SUCCESS in 1h 38m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/34660462b9244bac96d619af20e36d8b : SUCCESS in 1h 30m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3bffeaffddd4505b4d54634cf74ad47 : SUCCESS in 24m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f2c639fae037405b95b243df8431c2c6 : SUCCESS in 1h 08m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c71ea2114aed4d22b5e4d454f7a9c8ea : SUCCESS in 1h 23m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a02a9b4b68ee44f5a84cc9533537dc9a : SUCCESS in 35m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/febfc9157b5c423e84c2a6d4c704f529 : POST_FAILURE in 1h 53m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e30ec3655e7a4d2d87e4f4dbe7dc3d13 : SUCCESS in 1h 20m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3b72795177447e99c5080f912f3eb4e : SUCCESS in 1h 31m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9632bff18bb642f8ae9ccf9a2a494779 : SUCCESS in 1h 08m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/446fca6a7b7248a695bd174460fb0ea3 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1f81ed6e9b73eb56828770a95fb2e14d26869164","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 14:56:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b0cddb9c466251248def650a37195c25b28ba3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 15:48:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b36c35c465544ce9e78956b2033bf24\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9d8dce8d36a4fd3ab5398597ab284d3 : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/676e39a7877d459c8b914c46834e1097 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7972174889cb4a2bbfdef0e1d0f2ea78 : SUCCESS in 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c4992f7fa5bb10a659639e91d55f6849ede24fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-25 16:28:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4a638e56b0cb4e4792914120a717e6c1 : SUCCESS in 1h 15m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8feac9d4beb34f589ca520ec91cabf7b : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c23d0b592282bd990cd139ebbd3f73225c4d4d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 17:18:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00efe8eac3ad494fa31c1a18175d143f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7313b054752a4ee98ef5a6c543e8830c : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae95d0d5172410682762f19b581e369 : SUCCESS in 14m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e07fe3e9f1d44f398b645c8edf425578 : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b2fcf1a6a704394b02090d86a9e4b65 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d71ff2c1014796b7cef3f3df649b90 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6942936d0d0b4d5abcf8aba8669b5988 : SUCCESS in 42m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3b38ebb778149e09839542477465cc2 : SUCCESS in 53m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c87b8c41e3c04c33ac187807ec694e1b : SUCCESS in 1h 15m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3bd244306b04f8d9c5d8784c817f1bb : SUCCESS in 56m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f25b32d2f0804a2e8720d4c2c5f8b3bd : SUCCESS in 21m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9f88fe233d449b199bd54a16551276e : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/90b1bb11d9224731b711981afecd9f7b : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7aad945c637d4ee9a8050ebeca89bc56 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/054ca10f1590463a85dd493fc5dff0e5 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1cc62105165347499edbbfa399552b5e : SUCCESS in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37a75b1190d44da09c8dd12d75d885f5 : FAILURE in 35m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02d7060156f24eecaaa41416375cb222 : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4bb434323d0e46a6bfb690774a1dadd0 : FAILURE in 1h 51m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1172b5cbc88a410889fb040dc9be8bb8 : FAILURE in 2h 17m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a943bf7d384c42899de4a5456a096d63 : SUCCESS in 1h 48m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/52919bd63a974795aa819eebc38fbfe4 : SUCCESS in 1h 53m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d03987f21bac4b29a1c5947054287012 : SUCCESS in 36m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b7d744585344bf5835409c378765912 : SUCCESS in 1h 01m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e6a1f54bd8174c6e941afe4b890d6783 : SUCCESS in 1h 18m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/68e90e7838684f7e8b28d85aa9a97b25 : SUCCESS in 40m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/039154f717c846cab83115aff4c5e8bc : SUCCESS in 1h 26m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5f1e12610a0e478faeecc357de22a89b : SUCCESS in 1h 40m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c35fe5c85d1047249231951d28b05013 : SUCCESS in 1h 29m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e91ede26c1a44bae905dfc0eb18174b0 : FAILURE in 1h 24m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21f573f54f6e4325b42a37e4824c1aa2 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4baafbcd3a1d9f570155587ffa7726337b8e7195","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-28 13:07:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0efc237e9912f24e0b9661911992d87ee53e5a2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 13:38:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63cf1afe9ad84b509c2361eba6c3062f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a217179e14684d2c86787090e5fd64a3 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c41356ab8124272a0d7be44d6eae460 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ed553def9f543fa938b0b938217a5e5 : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f58f00923f5abe056e80a418fac7ad82b47c07c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-28 14:32:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c2de7f7ba1b41f5b8a0e0b09cf08aa7 : SUCCESS in 1h 18m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/423a1d3d531e47c2be653799eb3426d8 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ff11f7f52739c4a058b66db9dfda5ad02ec1ce32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 17:12:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0195b9823254e229a5e6791a95717f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3441f313ee3740f9958cb23bd800df46 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8d196c0fdb441a3a3629791ead472f5 : SUCCESS in 12m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2236eb9611254ac886740786089d3772 : SUCCESS in 13m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0b15fa7c2aa4629aea3154aff6bab34 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcf7e89757d4901be18c3344dda881e : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/746dfa677d624deb8a46650f15ee0bf2 : SUCCESS in 52m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6af16bea85dc450c88cea8ee08239910 : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6b1b34d4a624a479f17c46723e3a847 : SUCCESS in 1h 39m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e2f51e176334653abab9d78bdc44524 : SUCCESS in 1h 03m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e831861aae7442cfbc76e0811b6bd001 : SUCCESS in 19m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/64fd58190f3e40b394cc7ab8a95ef071 : TIMED_OUT in 40m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a329df45a1494f4a99dcf7e5e239ffca : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9578175215aa4671b79a2e8d0861058d : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a73d3e5359f24b5ca30e1a98f7d5a6d0 : FAILURE in 4m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d4ec3ea38c343dbafd8c2c48968c9ac : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b63f1e2df26e4c19bd96daba10cf2f6f : FAILURE in 32m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd383e464bc64c3a9e7436e92eb5fb6d : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c27664b2d4c74bbaa1abaec1866167a3 : TIMED_OUT in 3h 08m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31c4365ba8c84858a0537f1bd74cedbb : FAILURE in 2h 59m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/48f37af3486447a4bbc4a7e0e8df2f84 : FAILURE in 2h 46m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fe45f2ec16f84a39af937e2ddd572979 : FAILURE in 19m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/abed5a7ccca74312bf2bf9c889d80e1a : SUCCESS in 38m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7a0583b0d1ab4699b9abf31e41373ece : SUCCESS in 1h 22m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/23856b97b44448dc847a1944e6f1ea0f : SUCCESS in 1h 37m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6a8e5b0298e34b758347c587525fb1e4 : SUCCESS in 36m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/700e043d445e4e6da7152462ee050ff0 : SUCCESS in 1h 29m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f791c74ff9ef4933b04f337b49d72927 : FAILURE in 21m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c7151bb3bc14f1d9006c715d79f892a : SUCCESS in 1h 23m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e7d4f3955b40639a500cc53f61ad62 : FAILURE in 1h 22m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/940788f3d4e04b70aaa8c4d72c474751 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"76af220734d5521e3c54e5c92f667a5232e33996","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-10 14:00:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f94965642c20eca9755f0013d1d61253decefb6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 14:25:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/211840a78c80439f8014cde89871cc15\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c13a540606564a1dbe14e0e623e868f6 : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd7785089d0a495a82aee764e4ec4df8 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bf57456479343249f3b9310c994a520 : SUCCESS in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f13f8d0b2355754b57fe4da1558c6fd9df11d7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-10 16:04:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/029983c390bd4751884eb7b266192051 : SUCCESS in 1h 20m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/832c1ab52a3f443b88e0f26ea6fb5cf6 : FAILURE in 1h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"22474635ff85dd4bdb7a3e2c8df4a2b7f22f868d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:57:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d4c68d94724d5e82ef700b2b7b90ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a2582af4c341d49f8f2e6832ce310e : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6da71563d76e44a487bf919d67ac88c5 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c1cc4f7f35248e68af7009b79ea02f3 : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa403b191ff14a78a85dc3426b17a79a : SUCCESS in 15m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c8c0bf7da64bdfb44b029780bd31d6 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d76073b7848f45099b290502e7e8da5e : SUCCESS in 29m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c78856ae25e40a9b611ee4fb3e73c8a : SUCCESS in 58m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca8814a4db8d488ba534e4460b1ba55a : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c5b46cd3ca24da48f9adc80428f997b : SUCCESS in 38m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa0edd25b2b34cf1836af2fc82cb06c1 : SUCCESS in 17m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c6a12bef30564204ab9cc504467005a2 : SUCCESS in 24m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fc39b0f0dfb41c2b2a22c00d1751cda : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b906ccc0971c48608dd16b951afd25b0 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50348320efc84addb2b7ad1284400f8d : SUCCESS in 5m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a15a9eae4903489eba3315acd02df4e5 : SUCCESS in 9m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e4ec23963cb84ca1bccec1e2a45e0b49 : FAILURE in 35m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca41b4a53d3249eaaf58f7b93b0ac14d : SUCCESS in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab3a2c68f6064f9cba69a18a38a6d88b : SUCCESS in 2h 02m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/61a7e2816c274385a7df781175d7c00d : SUCCESS in 2h 43m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61e8dc25846d4effbe83c6856a2b027f : SUCCESS in 1h 54m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b1a7bc36ab64046a36612e02272f046 : FAILURE in 19m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cc87377e81d44f92b97c26a9d2fb5061 : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad789e04efc247bf984d44a4a0271e6b : SUCCESS in 1h 12m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22088ef654714ed5b47d5745928a8fcb : SUCCESS in 1h 48m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/79ca4025d93c4526afc168603b352cbc : SUCCESS in 37m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46830234fcaa4867b2b5db913b8d4449 : SUCCESS in 1h 14m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/688f8488c2204762879f961d7f177928 : FAILURE in 19m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb828acec6514520b843a16612ae293a : SUCCESS in 1h 48m 07s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c14c16821364ed3a4c45f43f02bd730 : SUCCESS in 1h 36m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/494959b1b1b5447bb82b1a86bde85ae5 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c2ced2e9cfd9e7dfadd93107ce8feb47a0a69aae","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 16:04:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b8435d8789cd5691126b86dbf6bcd3551001bbf6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-12 17:35:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06a0644530c6e3405746daddad22ff6c3bfc5f54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:36:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a07311f5c1d6320c321176847913812ee92e47b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:36:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebd90964b71141fbb9b4ba61bcadf460\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e04e3ee6ac472c9641d7994e552c0a : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93a963d09e5145dd8776c6af7ef083d5 : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59cb54a7b2af4d31aecd89e1d66c6a04 : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61084b322594bb9954349155cb92916 : SUCCESS in 14m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/0fc1ecae8d0d478b814b91082a502cb6 : SUCCESS in 57m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d148fc1fef9d449583f5e068e984468f : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2c6d519ef514d62a03a25827b1fe27a : SUCCESS in 1h 30m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e48818ea9ac34f8b85827d05fdda87d5 : SUCCESS in 1h 08m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38d40a342ad4a7ca2c37305fc85ac42 : SUCCESS in 17m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb07e3c92b394b3aa9b757292ef1a4de : SUCCESS in 1h 12m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11472708eb8d4ca3b77a44478dc142f4 : SUCCESS in 1h 59m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5d800516cb49abad97c81946598651 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f6fbb9d5078d705982235308bbcf71121a38d99f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:36:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"37ac2fbca83082fc41ab4806ad9c4f6ccbee1710","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:37:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83f559c59cfe4162b08aeeb6709d3e9c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911cbc045e9e49ecb4a7776d8440f01e : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6affa49a3e794c518fc113ffc1293a9d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"5b7bcd00c699879d1f79dc184cbdeb3d0f208088","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-13 06:54:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-946858-2-fc : SUCCESS in 1h 2m 20s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-946858-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd9c9e8c11ba0e58bc3d855933194438fec7f3de","revisions":{"4219bffa5c3851709139ac5792b5f2c752e691f8":{"kind":"REWORK","_number":1,"created":"2025-04-10 01:30:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/58/946858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/946858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/946858/1"}}},"commit":{"parents":[{"commit":"2b1801f1e5caabb91bcd3235ab3ff1d1a7fc43a7","subject":"Clean up volume_opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b1801f1e5caabb91bcd3235ab3ff1d1a7fc43a7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-10 01:23:24.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-10 01:28:05.000000000","tz":-240},"subject":"Handle volume_clear_* opts consistently","message":"Handle volume_clear_* opts consistently\n\nThere are 3 volume_clear_* options for lvm thick volumes, but only 2\nof them are currently read from a backend-specific section of the\ncinder conf file.  So an operator could define all 3 for a particular\nbackend, but one of them would still use the setting from the DEFAULT\nsection of the config file, which would be difficult to troubleshoot.\nIt would be better if all 3 behaved the same way.\n\nAlso enhance one of the unit tests related to the volume_clear feature.\n\nChange-Id: I623ca7f9061b3e3edb9660da65bd9d64309f52d4\nCloses-bug: #2106680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4219bffa5c3851709139ac5792b5f2c752e691f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4219bffa5c3851709139ac5792b5f2c752e691f8"}]},"branch":"refs/heads/master"},"cd9c9e8c11ba0e58bc3d855933194438fec7f3de":{"kind":"REWORK","_number":2,"created":"2025-04-17 20:01:54.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/58/946858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/946858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/946858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/946858/2"}}},"commit":{"parents":[{"commit":"d52e0532f978507a742504d4ab1dec3fa8473b4e","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52e0532f978507a742504d4ab1dec3fa8473b4e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-10 01:23:24.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-17 20:00:10.000000000","tz":-240},"subject":"Handle volume_clear_* opts consistently","message":"Handle volume_clear_* opts consistently\n\nThere are 3 volume_clear_* options for lvm thick volumes, but only 2\nof them are currently read from a backend-specific section of the\ncinder conf file.  So an operator could define all 3 for a particular\nbackend, but one of them would still use the setting from the DEFAULT\nsection of the config file, which would be difficult to troubleshoot.\nIt would be better if all 3 behaved the same way.\n\nAlso enhance one of the unit tests related to the volume_clear feature.\n\nCloses-bug: #2106680\nChange-Id: I623ca7f9061b3e3edb9660da65bd9d64309f52d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd9c9e8c11ba0e58bc3d855933194438fec7f3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd9c9e8c11ba0e58bc3d855933194438fec7f3de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
