)]}'
{"id":"openstack%2Fcinder~946912","triplet_id":"openstack%2Fcinder~stable%2F2024.2~Icfc28c884a44ec5124261de71061c38ed1d13679","project":"openstack/cinder","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-11-03 18:14:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfc28c884a44ec5124261de71061c38ed1d13679","subject":"zuul: cinder-plugin-ceph-tempest: raise swap size","status":"MERGED","created":"2025-04-10 14:17:40.000000000","updated":"2025-11-03 18:23:09.000000000","submitted":"2025-11-03 18:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946912","meta_rev_id":"30099b852af2028469d273c20f54e0e9beadd376","_number":946912,"virtual_id_number":946912,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-03 18:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-22 19:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-22 12:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-22 19:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-10 14:17:47.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-04-10 14:18:30.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-10 14:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-10 15:36:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-10 15:47:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-10 16:03:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-10 16:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 16:10:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-10-22 12:28:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-22 19:19:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-03 16:45:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-03 18:00:59.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"ba7471123f3b87c4403fc032891312d7648f8e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-10 14:17:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d596aed1b562f0dae8bf7d19b5a356b21deb567f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-10 14:17:47.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1674/","accounts_in_message":[],"_revision_number":1},{"id":"173fbf4e9082f793078b252cc9b97ebe31a3b690","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-10 14:17:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946912/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946912/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"c0295c37e3f805f4cdf5cce9f29730f245a09151","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-10 14:18:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1fede639042867f56ff5e41e45e76c8aa816fcdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 14:30:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f40e6a3514445ee93ca09dcb5c4765f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d64bd528a3e4cbc8ac014de86e3f23d : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b059df0cb4845a2a2e898b2ae2ef8ba : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8f8ad3bae6346fdf13064cd49020908f66b6c02","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-10 15:36:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946912 : FAILURE in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4987d34a73690c3e7305b137d7d671952f652b35","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-10 15:47:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946912 : FAILURE in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9fdfef95270821050ef753f9942921cec363ea59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-10 16:03:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ede9bf26a02049c19e4e325b05142255\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1baae09a17b484cb54b7ae109395219 : SUCCESS in 47m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/06d649f0e4894bbbae3657eaa2eb302d : SUCCESS in 42m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c0f98e044c5495387b2cd212f19d4e6 : FAILURE in 1h 44m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7d7b68abb5e4ba99d93d9705c98ee7c : SUCCESS in 37m 04s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"e42160161a1aa370cbbc12b8856ebebb4e31676c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 16:08:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4626dfb91ac64a68af54df988b05cdee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cfd74d5ef6416eb00afadc89813a8f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/844c611b2ed742a58bc86d7f942a333f : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbe5c9febcaa41128a8c6de4a14b077f : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b77acc6c30484d8392b3148bb0c025c2 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04eb9b68b512478cb3c27d23282402ed : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8e49c7d336034433a9ffd6514a629e36 : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b86995fc86a42b29e2957aac7fd081e : SUCCESS in 1h 30m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/226f9675f24640df8471f049e789d328 : SUCCESS in 57m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e2ec66a7d1884051af6258b06a26e167 : SUCCESS in 21m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9dd0caba9057414ab4e55b5c5b639f09 : SUCCESS in 3m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06f5acb64f28404983f3ae29ab824ceb : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29b20c0f687a41d891bf94816b9cf804 : SUCCESS in 4m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0fb5d14440804bfeb6d588d2817189ee : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6729d982ba72429f934d75d0064aa981 : FAILURE in 34m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c4d8ce6856643d1900446cdc7ab7b1d : SUCCESS in 3m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ecadd17a7d3a48baba8006329c07b685 : SUCCESS in 1h 33m 47s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66ca0f5339314dd4b246a26c06053b6d : SUCCESS in 1h 41m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/112b790f8efb4e688809be2a96d35ef2 : SUCCESS in 1h 49m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/72ebdc84354b46d1b9b24ea6257c1550 : SUCCESS in 29m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/20306e849cc84e178ade93abcd080b10 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17336a9e52814f7a9efeca9efd1fc84c : SUCCESS in 1h 35m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7d84a04926643a99b488d5565923920 : SUCCESS in 34m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e86b7cb754a48c488649f84913cd7fd : SUCCESS in 1h 24m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2df60337a76b465684d0561cb1e6964e : SUCCESS in 1h 25m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26f7424381a6416d9f9c037ca8c8fa37 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3eb5334a5434a6f51f01435096e1c7a6acfc45c4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-10 16:10:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"230c3a042196da70362738b514c5631c6e39c601","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 12:28:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c70e3e655449077f2cee4d3415f9cb37cb34e5c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-22 19:19:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"063d8bfa5d6ca9fad884d5a0a6f3ba2d956d6c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 19:20:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08b14a2545771a2bc2e8891652e430c0fedecf83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 21:30:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43af5f4b2ed642e784c674fb8a58dd70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42ddb2fbdab4be298cbba657cfc7cab : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e5c09248d83480b82ff01c00cb90a41 : SUCCESS in 11m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe4d2486bd77454b848414713d3cb464 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ae306524484545b700bd0e27d80e1a : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/04e4ff8ea07946d2b4620f615f4d7447 : SUCCESS in 26m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4fb0c3914014e379d98b97c1f066d2d : FAILURE in 1h 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60c32c8635b040d9b83b99e72eb01398 : FAILURE in 31m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9eebf11aede44b8c888eb62e69c9a50a : SUCCESS in 49m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/afb2dbcf9c3f4a0aa36eff6789e8a478 : SUCCESS in 1h 09m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddd5e002a73648ce8ae8f804d8a327a8 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":1},{"id":"818a953a59c112e7d01fc4b4d70a98e5378d6d8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-22 21:38:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50b3b13b3c8dd70a2e646d2b1227ed76d0ad7131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:41:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b557735254d479899e25b374062ad0f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/014a0bcecc7249e5b37a67edec16abe3 : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb378b7e1c924c1ebac090002ce5756f : SUCCESS in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22ebb8675ee3d9db042e311083502782db6d1679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 00:35:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc023e73a31148a2987e863e484ec023\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfb1dcc41f846d5ab9a347e8658a0f5 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/897733db93694637a88902780b0b359e : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ee780c229304cad9c4338a41676ffff : SUCCESS in 13m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1d88da6260d4297b644860404ebc75b : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d04f28fddba4c1b9244f3597cdfbf66 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/2d2606eff25547debab1fdda4365b5e0 : SUCCESS in 55m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f0f3cbc50cb48f4bd60cf793110842d : SUCCESS in 1h 23m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7974edbfd1ec46e1a33c17968d9ded6d : SUCCESS in 58m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e994aa3445ca42e09e9aef0b42fcbed3 : SUCCESS in 10m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c34e53081efb49eaa21bdde8c877fb18 : SUCCESS in 3m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f72cc3aaa50a476a9a614c04f05c2bcd : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3492dce69f2b4502b273a102546d036c : SUCCESS in 5m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d0ba172966c4e3a9114b0b71a6a0708 : SUCCESS in 5m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50be921bd55e4642a676abf0bd748c15 : FAILURE in 27m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74a48c345ee343e8bb37eb75f7a08595 : SUCCESS in 3m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/042391d4f9484bb1823bd285b8727b91 : SUCCESS in 1h 19m 29s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1e842e42f794471b145f71c677c0bb0 : FAILURE in 1h 53m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/83f3488f767a4d06bb28f10f8482048f : SUCCESS in 1h 36m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cae33bc5df064248b323248e3d1a7497 : SUCCESS in 23m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0fa541089d3438e84f29a54c9c2c05c : SUCCESS in 50m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/265c1375d8534f97badfc81e6f51b581 : SUCCESS in 1h 24m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81823290bb454400b10b385e25b7dbdb : SUCCESS in 42m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8d086f3d73444803b6069cd3f82c6912 : SUCCESS in 1h 00m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8dece9b11a664de5a485ea04f7620241 : POST_FAILURE in 2h 51m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5aa85dc229c481798e7ede447dbbebd : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d0118a0979e06d7043812628b116e988c040204a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-23 04:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44f84c25517b0b0a8743d315d8fd8a6f4c49e4e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 04:58:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58b9bfdbc4445bdaf700ef0ef8b260e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84f81771a86f49039012f7e411bc427b : SUCCESS in 19m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dbd1ad24771b405ab5ccc0930f40c4ca : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7431538f9ab0190820277ee7c7eb0697faf00987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 07:04:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88a2b6c8ecf34fde980d8b36b0807ee3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed019b7ed32b4fe29f502ba475da70fc : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/714a70647d684ed5a434e416c9b3ae9d : SUCCESS in 9m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/060f0ab300544def9d5c2c0f9253f474 : SUCCESS in 12m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e17e36598054599b42dd2583d819fa8 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a6479561f843c58800e22db674b352 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/948e176649cb42af9d198e4a75d291af : SUCCESS in 26m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31e3cc95e3a34c07a82941d782ac749b : FAILURE in 58m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7eae318f858c4ae2a8a1ef28bdba008e : SUCCESS in 59m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d05dee1e32354236b1ed00c3ead48720 : SUCCESS in 16m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7491fad084f94b24bc5d191c52e546a5 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c6f844f54cb4b11977422f1043b9728 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25f428306c164e79af1cddd6ffae8f71 : SUCCESS in 3m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d936098b7c354078b05e2439419bb057 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/054f0b8569104bb580f9d2ba421a4ab4 : FAILURE in 32m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b79ad0b8794469aa5c0259c60061737 : SUCCESS in 8m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e403b33607874d428bb8f01c247ce845 : SUCCESS in 1h 31m 45s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b299040e5b01436cbe6b651d1fe967d6 : SUCCESS in 2h 02m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a1c5db03989a4808a09770ad04550b0e : SUCCESS in 1h 58m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b7432d104ea34610bf386de610ddb8ab : SUCCESS in 33m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c396f0dea0a4545849f45e4d6e779e8 : SUCCESS in 44m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f5d5eec42c654441abafd34b9243335a : SUCCESS in 1h 14m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/837b96acff6847d3b3c1433ef5a4a193 : SUCCESS in 42m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/027989dac0aa4b47af5a8a518947e703 : SUCCESS in 1h 08m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8b796d90862c46faa2c1a47915e6d6f3 : POST_FAILURE in 2h 23m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/842ce77d9ef94f229db1468b4fed95cd : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"efa72ef100623e84eadaf7127cf24325e8003628","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-23 14:28:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bb7011aac676691b1120a9637fd8c6c966e8f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 16:36:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/880591ccece048c3a7cf086fe107c656\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94af58d198d1431fb6991aed48d6005f : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0075fc3b2d9e4ffcbbce4b1412df142f : SUCCESS in 14m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ede8c5c1e434287af83b8c55bdc74af : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12d328223a9d44e98da59748e7bb033a : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7715af33813940b1a055108f8d571970 : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/69666fefd5cc40c2a4b703ad84b5d293 : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d09397713ae409e88823dc995efe379 : SUCCESS in 1h 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8f1d783f4cf4be09d0cfbca31e21f23 : SUCCESS in 1h 01m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd7b2f078f5a4f1bb6ad01667193454b : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4cd4f7bcf0ed45098aca21b806b07e29 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45f2b1ab362347de823c6b4d11800c12 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/483571ac555f4e048f4425598f9fd45a : SUCCESS in 5m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/705fa20e62ae472f99aa362b8f6fb831 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/251619014bdc4f2885356211d9dea87d : FAILURE in 35m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e599d176685e4e1ea78ee6b31884e5cd : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/03b462981d3c455fa2335f44eeda5b75 : SUCCESS in 1h 38m 33s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4dd6baa7b92410aa44e4241e520cd92 : SUCCESS in 1h 28m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d7719ab8805b4251996b84a44a3f730c : SUCCESS in 1h 45m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1183ccf72a0240f6bb4f2e0473508d1c : SUCCESS in 30m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5f872c3835474d1483bfe1a36db5d0dd : SUCCESS in 1h 13m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/66e219dfbe1640559efcbf23463715c4 : SUCCESS in 1h 25m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b0b5a6edee114ff596ef36b3921e940b : SUCCESS in 29m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5aa65fd5eee412382fde2b2cc1b53ca : FAILURE in 1h 00m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cfcdf9b7a13e4481bb0d956a1126eeaa : SUCCESS in 1h 25m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5df8f259da0148e2be9ac06974429b24 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4431ac331507087bd411dfa2b3a7c6165a4c4cb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 17:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6d8d3029f3547c1a05be623ca6cd96c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6377e5f01284042bf8bbcb9f75b0565 : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8cc027df8cb5481f8a19719178dd40c4 : SUCCESS in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f4e96f677c5cca4d31738ed1dfaa341427d1af3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-28 14:46:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b62dabda6374226006351392758c85e177c8308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 15:37:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e80d86c45da849cc830bb1f9d63ca6be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/596e7fb57b5142b0bbb7d8ed461f4dc2 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b50f1b14df6a416381d071742f1a62bd : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43049ec1aef7a3ddda0365284fe4e607aba1fd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 16:01:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/081f42e633cb4b2d92f7a7c6220a4baf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548ef42929904352b178ee3638037c68 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8f32eaa76f4409489b78279b99a1c8b : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c005b99508b243308a7a2a7e2b7f6e2a : SUCCESS in 7m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34c1310616e34b5ab84e5d9635bf0952 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58203921d70444d949e578869a6d032 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/51ccf913e61c411a9ecb9f90ada9b9d0 : SUCCESS in 56m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/299dc4ddb530430b8058035d205d2cfc : FAILURE in 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fadaeac85e614d75b4967e71f2c0ad15 : POST_FAILURE in 11m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d6c7b4c7ccfc491cbe5adc80b5cae387 : SUCCESS in 17m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/920601536ab94602901e1bf9ce806648 : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a0aa812952942beb46d08a189ee0e24 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/397ea8cd4dd94144b329924c30284698 : SUCCESS in 6m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6dbe3a378de040e8a0a5c114d6861dac : SUCCESS in 9m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b6a70946b4bb48089e221d12df03ae53 : FAILURE in 18m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2023777c9c92477492bd62108ab05ff9 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3598262d06714b1c8bc46d0d6736056b : FAILURE in 18m 11s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e280d4ef500047d08d7a4fdc7be053af : FAILURE in 18m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/54f11c61aa5848c6bb6486ab8f2b524a : FAILURE in 39m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b9450a8f19c4d0a928f08cd336d44ec : FAILURE in 12m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24be18efc87b44f385a06656368ff3f7 : SUCCESS in 1h 08m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f97ae67a8b3a49048e68a0ea385f30b2 : FAILURE in 8m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6018b1902276476e900931bfd1bca0de : FAILURE in 18m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dfcdd04d12484e2388b5f68bd6a06c19 : FAILURE in 18m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e2dbe960b9ea4c6191ee1756ce1bc9d5 : FAILURE in 38m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0b5c400831042978fbd61941a231519 : FAILURE in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d7d736fd6d8aa231c895c7ec9b7a449b6f88b51f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-11-03 14:44:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bd9492b5e487f9c04ef36c526ef4ed227f4ef57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 15:04:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c92c9358bb4cfc8726e4a5e8f6d763\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a198172786124503833d0e93a7de76b8 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5dff02d6c2a441e5b205d805120af3a1 : SUCCESS in 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50e52748a63e5223148c6a264bb3f8c7e53849ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc5f9351a9d4d4e9dcad120fe64b04f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02de8793a5b40e08f74dd7713c089b8 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fdfad7489db4f3bbc297dfe4d276de0 : SUCCESS in 15m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5429b58b16d3493fb6e99df9f3b65487 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0248a622c1de470fa06b8f80fe35a837 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a76063a1d9546d0b3a9cf60e9725bdd : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/34d54498d72a4445884edd63a217694e : SUCCESS in 29m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/329dfc4760db42cb88d532038227f3e2 : SUCCESS in 1h 30m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d59e7980d4af40feab0e71f23134c3f9 : SUCCESS in 58m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dba47673de144488a81fbc52da0dc510 : SUCCESS in 18m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/52b677b5557e45078f13eabd656107ab : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80f5037a4c7b4f4989285999c7cf9868 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1cca1e2a61d4b4f9488f36c116d4858 : SUCCESS in 9m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/01a62e99159647499cbfe6482d93162c : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4eae8d40c07e4cc1a60132b968f98043 : FAILURE in 33m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a3336eaf132410fa576cf49fe57bade : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d0d07babc811461693cbaf2f4c23b600 : SUCCESS in 1h 25m 31s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2cc6befc4228420bbad0f2eb9f29d83a : SUCCESS in 1h 53m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4b7808cd3d1848219809fe7719691a66 : FAILURE in 34m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cf7b7c551eae4d3483282a65918cf585 : SUCCESS in 33m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb156cef6a194889b7fb7fef711e5dfb : SUCCESS in 1h 08m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2964213b3f3941ba984a72f5f11f6a7a : FAILURE in 1h 24m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92186c9651f84382b60b4e3a5641bc8a : SUCCESS in 52m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e37e004b68204a6a81913ef507f1dc0d : SUCCESS in 1h 11m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/52d219ea7bb14677b0c5b24e9d3c6835 : FAILURE in 18m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/217d140e5ce3493fa1e7f2259eb64d2f : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"62695b4eaf23d5d864f1cb25353008a1905c98eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:39:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaf72e4c2979ee6c87ebc0d16071901e4e138b13","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-03 16:45:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_11_2025_16_39_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"d5bcb0ea72a400fdd1501e03f8ee01870893db31","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-03 18:00:59.000000000","message":"Patch Set 1:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-12-946912-1-fc: NOT_RUN\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-12-946912-1-iscsi: NOT_RUN\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":1},{"id":"1fc83d1d60c6ee1895cdeb3715df344e1e83dbe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:14:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b409af081f4557bf05ad4a45172e6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524ec944a9b74598a0311bf4e83bac18 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da31ae5044c041d09382bb3792bd5624 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dfc2f17c5814a5db6c4d7dadc36c067 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e494a99409ad4a7299a3301941166b04 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/25586ac9f4d14785ac0342b963c9db93 : SUCCESS in 40m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d31d93720c114bec9ba36772bf98dd86 : SUCCESS in 1h 34m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c170aa381367475f820341da5aee68e2 : SUCCESS in 56m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7061d920537b40c6b7651de7c6ce9f06 : SUCCESS in 50m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7fd5f659331848dbac42a081adbfd56d : SUCCESS in 58m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13429b265e9a4f7ba160e7e85ab45a9f : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2d689a5966fd20167b548eba64a48dbedf5d47ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:14:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50037ff5654df815d581569512d926eb2a50f4f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:16:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee77b6e5e0094587ae3f346e4b9e41e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68704880659d426795ea0593b724e193 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"30099b852af2028469d273c20f54e0e9beadd376","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-03 18:23:09.000000000","message":"Patch Set 1:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-12-946912-1-fc: NOT_RUN\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-12-946912-1-iscsi: NOT_RUN\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36e7d28cfd8db61301c76b3fad49b329f2718d1f","revisions":{"36e7d28cfd8db61301c76b3fad49b329f2718d1f":{"kind":"REWORK","_number":1,"created":"2025-04-10 14:17:40.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/12/946912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/946912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/946912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/946912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/946912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/946912/1"}}},"commit":{"parents":[{"commit":"c579760ddaee4e65dc45f8f546814763af0da0ef","subject":"RBD: Fix upload volume with different format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c579760ddaee4e65dc45f8f546814763af0da0ef"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-27 14:48:57.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-10 14:17:02.000000000","tz":-240},"subject":"zuul: cinder-plugin-ceph-tempest: raise swap size","message":"zuul: cinder-plugin-ceph-tempest: raise swap size\n\nThis is set to 8GB in devstack-plugin-ceph, unset\nthis here so that takes effect.\n\nChange-Id: Icfc28c884a44ec5124261de71061c38ed1d13679\n(cherry picked from commit b05608e30399df92197962efef72ffe8de9794a3)\n(cherry picked from commit 2889ee1299a9db75e7acbe4cbfb4601a725741a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36e7d28cfd8db61301c76b3fad49b329f2718d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36e7d28cfd8db61301c76b3fad49b329f2718d1f"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
