)]}'
{"id":"openstack%2Fcinder~946995","triplet_id":"openstack%2Fcinder~master~I49404e87eb0c77b4ed92918404f86c073fbfd713","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-04-21 17:08:38.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-04-21 17:08:38.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-04-21 15:30:53.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-04-21 17:08:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49404e87eb0c77b4ed92918404f86c073fbfd713","subject":"NFS driver: Fix fail creating volume with multiple snapshots","status":"MERGED","created":"2025-04-11 13:07:19.000000000","updated":"2025-04-23 21:23:51.000000000","submitted":"2025-04-21 17:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"946995","meta_rev_id":"d6a6f3b74e8a5bd96a0951269ab0753c89936c5b","_number":946995,"virtual_id_number":946995,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-21 17:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2025-04-21 15:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-17 12:33:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2025-04-21 15:30:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-11 13:07:30.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-04-11 13:08:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-11 13:48:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-11 14:10:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-11 14:23:43.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-11 15:12:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-11 15:14:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-11 16:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-11 20:59:41.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-04-11 21:04:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-04-11 22:17:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-12 15:12:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-14 22:08:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-04-16 07:01:38.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-17 10:12:56.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-04-17 14:01:42.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-04-17 14:20:16.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-04-21 14:00:43.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-21 15:31:13.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-04-21 18:48:55.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"745403b59c943a1a4790a83c997f1c28806b8184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-11 13:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d6b8df5fe1e86344e8200869e4aec1b582b7684","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:07:30.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1681/","accounts_in_message":[],"_revision_number":1},{"id":"9d098de81a2ca3384567df5ed5f1728f5fa4c538","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:07:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"b1993bfc5412ce10b3cb928aeed8fdeb3c2903d6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-11 13:08:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"0d994a8ee01d69e5137b3708f99362e3ca8fd8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-11 13:12:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"13833abec720753475b9ad3f1bf84af63becc7cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:12:40.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1682/","accounts_in_message":[],"_revision_number":2},{"id":"c00161af49a611d9df9a3bd4fbdb0f1aee1293d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:12:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"9ac1fd22a1ab09ed34eaa75a32a49d5b93a00ec1","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-11 13:13:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1d515abeeafa1ef8fa562d7bc80726a02eb58d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-11 13:19:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a8126549287c353fb2ffee476f894cd82230bc7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:19:35.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1683/","accounts_in_message":[],"_revision_number":3},{"id":"73a5d7bb43098f7520718f8240b429686f58b727","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-11 13:19:38.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"560f4272e64707295a24f0d6cc7977a8f6fa31b0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-11 13:20:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c84c08f1d233015bddfda5b67569a64d4a8cf2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 13:48:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/871a25344a954f0c8bad630ea91e10d2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c7619770c20b4e938f98279dfbe7c4e3 : FAILURE in 12m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6966425e40c14e4396dd6da75afdd7f4 : FAILURE in 12m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb51195850c14771a8e5e96c2a7ec8a8 : FAILURE in 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"08ff26319f3dae5f873f63350a7a2832e7cc0a1f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-11 14:10:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946995 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":3},{"id":"5c7c071ab85430afab0f0134e5ac41d41d179b21","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-11 14:23:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946995 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":3},{"id":"554511fdd1b557bbd4a5b97745bece9bd1733b35","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-11 15:12:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/97c5ec2799454698bcaeac771573b21d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d22f5041eb4a4b04aaab53d65612caa1 : SUCCESS in 51m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e1764e88d994130b51909d996d3ea56 : SUCCESS in 49m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/caeba5d8068c40c6baab75a8b560cdcc : FAILURE in 1h 52m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2515edf0c0094c1ab5093cca79b30191 : SUCCESS in 40m 18s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/912ea52f4c7a4e97a0d6d097e506ae50 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7d47adf60224f68ada9d7973757dcc7 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5443a8a01142436480ac489177ef3bbc : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c70c2353f34f4c10a8d1e4d0bf69c63c : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"9c10e10bf4dd9871f5f6cd446d118f063dbebc48","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-11 15:14:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/50e8b2cba17c475289a6d3a53c0068b7 : FAILURE in 1h 53m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9c7ab0e419f646ae80c1507ddaffba6b : FAILURE in 1h 26m 55s","accounts_in_message":[],"_revision_number":3},{"id":"38ee1f572260ef771e5e3b6ee38d39b6db7b1ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 16:22:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7cf21708d4745a699ec71f0f80bd1de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe52e1a2c238408aaa66d7e88ae3a799 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9b15fbeff4e42cebca2129ab98cd8be : FAILURE in 9m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c13cee6a45824ddc901ad06f5dbfbc4d : FAILURE in 10m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d754634db7f8484c99223c1aca7cd833 : FAILURE in 13m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325685d8c15244c1992b5d64dc24e792 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b936f739b0914eaebf7cdfcc345a9398 : SUCCESS in 44m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8342a3fcebe44e47b1f45f380a721c1b : SUCCESS in 59m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/749cdfb85c7b4bebab5d0cef5b2d09e5 : FAILURE in 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21c0f1c2610f4b5eb309c02b50f786bc : SUCCESS in 1h 01m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34756d6e0ba8422198a63e15c19be8c8 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca20d1f4ebac4a8eac9814704f02c15c : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a0ddfe081ae43619b1dd6a0905de47e : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92c592904c71466a9a60fde62f3c9bb0 : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cb00589742d34a629fd4083af15decec : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b5bc68de841c426a8eb304542fa91a8d : FAILURE in 34m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ba64af294dc468aafd63ac90924e430 : SUCCESS in 4m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b6dd4b9a7344a87920a756a2516369e : SUCCESS in 1h 41m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50f39d2b31574f089b8775602f098591 : FAILURE in 1h 19m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e51aa2f50a994ade96fe4cb10ee84340 : SUCCESS in 1h 52m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7852dd573bee4a2dba15875ee16a8a79 : SUCCESS in 1h 55m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f0695f4830524eed8a2f0ef5ac586b8d : SUCCESS in 34m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7dc1036862a14bb3bc76109380099467 : SUCCESS in 1h 18m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0774af52934f4e97a7737b1f37d81996 : SUCCESS in 1h 26m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/44ea0f217dcc440eae18144e0fcc251d : SUCCESS in 34m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4bd515375c674d4cb6f17a67199a2c51 : SUCCESS in 1h 29m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4aae92f625304bb4a60b22eea70be68b : SUCCESS in 1h 34m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ac0b4d3f740407496fc73d979891101 : SUCCESS in 1h 47m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6fbe60904c24df8a7383d58f2f9c512 : SUCCESS in 1h 34m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0fd37903ef64e709867cc05b2e7a33c : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4ad137c6ce684d668232eb6d7d3777b5e887c81f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-11 20:59:41.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-2/FC/ : FAILURE in 7h 46m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-2/iSCSI/ : SUCCESS in 2h 46m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"90d2f7d4f78efa762052ed30b2b7c8f4d56191ea","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-11 21:04:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/3/20250412021051/NEC-V-FC : SUCCESS in 3h 51m 13s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/3/20250412021051/NEC-V-ISCSI : SUCCESS in 3h 53m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9788d06a05fa63e25cf419ef7c4ac0767f705b95","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-11 21:27:08.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-1/FC/ : FAILURE in 8h 19m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-1/iSCSI/ : SUCCESS in 3h 12m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"10025c2d28f337627c8fbce885660467681890cb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-11 22:17:03.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"988506df93d03c8d17a06aa0c537310c2169c42d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-11 23:46:56.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9de5ff768bbd7500846e3d5be41c161ea83000a8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-12 15:12:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/3/20250412153855/NEC-FC : SUCCESS in 5h 56m 15s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/3/20250412153855/NEC-ISCSI : SUCCESS in 8h 33m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b103819ff2ceaecf599ce40853af0dfcab0aa3d8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-14 22:08:15.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0808f87ec779ff69ff2e515c91564bb5ad7508a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-15 17:00:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"04679096b8aab64933bd7eeda5ce2665caf4fd9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-15 17:00:37.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1714/","accounts_in_message":[],"_revision_number":4},{"id":"9cd0fc79524ff015324e3aa093ec3b70632b090d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-15 17:00:40.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"d40503af98c3201243ba343294d5e3b1de7c2798","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-15 17:01:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"b8d54dbb5399a7e317fb47c9a1a13c3f842e59fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:17:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f907ef06f944beb8227f4577a3ee5f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0baf7e66a01942079c5e048937ec1f90 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf47f1e1eb3d4de9987ae862710a97db : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4d672b713ab40d3b98bca842cd2b8e3 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f36ef84f61e9118ae9894b7caf2d47e7c5809b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-15 18:11:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946995 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":4},{"id":"277f0ff62500474bbd76b7f6af1e79b7cff1d9de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-15 18:28:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9a193f87a88a47da8bfda528cca90d2d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/27ef40b793584149a5a870c14ffc9bc1 : SUCCESS in 46m 28s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5d70fb918e2438f99f2b4f6257f6e9f : FAILURE in 1h 13m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3e0bd6040e6843e3b3c58594c031b561 : FAILURE in 1h 27m 10s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ccd08891bd54e33abea348e7f1d7716 : SUCCESS in 43m 47s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6f797be94334124a3cf245cff683247 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b94d15bdd1b8405684231143d91523df : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cdce41f8bd264622b2d7f8b24bd29fd6 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8ae63fc6720f49ae99871299fd2bed90 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob ... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"bb5099420b3816135f01a457af21221734de5d8e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-15 18:35:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946995 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":4},{"id":"ea5cf270e83ded760636514a63e2973c883644c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-15 18:55:37.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"efc58b4f5faeb0db9937d4d4b508d552a6de5ec8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-15 19:13:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e63bfe53c34741f28bf42e985aa4013f : SUCCESS in 1h 21m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/77701e0efedb4b75bc94b006e1075e75 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f8bbc42a257e50f606fee36e09659e0e33425af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 20:35:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a65ebed9f73c4877a2153b497eeebd89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418c8fc0d6da41078da8c9f98c65be21 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d22291e50eca46b7905dacc2cf69edb5 : SUCCESS in 7m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74017c69b0bf42cd97d6fb73959098cf : SUCCESS in 8m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8b2812b45244007bb41bfb6421e716f : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b3cad783b3420fa5a6f3e65251e8b7 : SUCCESS in 15m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c2a907b8ac00435e8947996787c1e972 : SUCCESS in 59m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ac131a45b0a4d75b8f5d5ef705e9a1d : SUCCESS in 55m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/719d1ece66544d78bd32fc09c6c1a751 : SUCCESS in 1h 48m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef35d37b127a409d90c708350fcfde2d : SUCCESS in 1h 05m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1875934b574431f9d46d4855e22dc82 : SUCCESS in 17m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b0023ba9ad7444c85cd8e2d2b565c02 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa4213ea84974882a5129601c513f858 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4685f8cff3bb4e4f9fef0d11e71cc16d : SUCCESS in 3m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a67cbd6ea6d64c2995797204e4363fbe : SUCCESS in 4m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9190b705e6c403eb921fc8553e329c1 : FAILURE in 19m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af372d30af234a37a38f1b9e92eca466 : SUCCESS in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1bcffe81cfe74aa0927391c2133c1872 : SUCCESS in 1h 11m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1ba4a2023ee74208a7e2dffe31bcdf18 : FAILURE in 2h 33m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/03ed9d8e00f94188adf324e43c6c979b : SUCCESS in 2h 01m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8afaf16f7e1a4d8a820688c8fa70baf0 : SUCCESS in 1h 27m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/28ead25b4eda48799bccbb286870ab3e : SUCCESS in 16m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6703b30f21cd41ef95591318d2f42257 : SUCCESS in 1h 05m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a0ea456f12d458ea44537fea05902f4 : SUCCESS in 1h 33m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/caedfa161e2649a48ed817b5fb23f79e : SUCCESS in 26m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef1b363add874cf7bec06a65c12b1604 : TIMED_OUT in 3h 33m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/63b176a376864dc3aa594a0e89edfefd : SUCCESS in 1h 23m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5be98519aa904db18583ab5c3b1dc77c : SUCCESS in 55m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f651a89e443240f289bd02641c407e3c : SUCCESS in 1h 30m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c9d3a8d09e446f94a7699ac58e0176 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6deede4c2662ddec33dfc0448bdad85c43b22b25","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-16 01:31:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5bb4e286257a6821c5262687fd1cc46d4a80ec32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 01:47:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f747e3281b4c487a80b6302d9d1f769b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39f57779ea9949429d933c26b330c410 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0d870f8f5aa451b82112dfaa9cf0e81 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/97701151ad61412e91ef049d3e0d5502 : SUCCESS in 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9f6f218fbbe4ea513c430c2e792feee98f0d3a4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-16 03:24:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/036a930ed44d4e8e991984c445c82102 : SUCCESS in 1h 07m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a5903dc4d4fc409aa37c3dc7c391a798 : RETRY_LIMIT in 47m 26s","accounts_in_message":[],"_revision_number":4},{"id":"0390a45ca1c4376e20577f46122ecfe70b2bffc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 04:00:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f32ca22d4248fab63e07f4edab8118\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f27de0bf554951935b19e372ccc4a5 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8812768e656a4e4b88c6334bae6bfa45 : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09057f3365d54b60ba4ac6a777188416 : SUCCESS in 12m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bfdad060fce413aa5666a2296c18337 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdabeaca6a24f47831f71b140de1526 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/03ece543b78f44dd999b801feb379cb1 : SUCCESS in 54m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/364d32b7afe340cebe010b51693a93ff : SUCCESS in 52m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e785273eb6cd49e4bd70860c5f93956e : SUCCESS in 1h 55m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c13d81a47f4e419f9980cfc19a44ebde : SUCCESS in 1h 03m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9cc8d3378ca24462993c8b3daabfcea4 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d4144abf78314d34bd271f8794fe6675 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2e590158d914b09aca1d475538177a1 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8468dd1beec49cdb003fb7db07401af : SUCCESS in 5m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbbabb30e0f64ddd89d2455602f9f833 : SUCCESS in 9m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0758803af7d643ec8e68ff3387435846 : FAILURE in 16m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52c74665b46b4b2baf847944d5f03cfb : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be75d7e108f349b79c9900278e71f2b1 : SUCCESS in 1h 38m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/91d89027544444a6b76b43f16282f91a : FAILURE in 2h 28m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19aff236d16f4c25b5e8c747d73c82a2 : SUCCESS in 1h 52m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c121a8ca65404787a0e349b9da60a4f2 : SUCCESS in 1h 54m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/43ad3b16f09f4c7999cf89a5f1e559a2 : SUCCESS in 16m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b56a2b700a564db68c31c407945af939 : SUCCESS in 50m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3542d812e8e9474893d685255d20eac2 : SUCCESS in 1h 35m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/58486c5734604ef8b7dd340f48b852a5 : SUCCESS in 40m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87aeb4dca3dc4d4bb4b83c3a75e2bfd5 : SUCCESS in 1h 24m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0ff87a7b3f6a497ba85e4a911410d366 : SUCCESS in 1h 56m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2f69760f4e7452db582b261e9ae73b9 : SUCCESS in 1h 35m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df9ca35dda9a418bb2fb828573faafbc : SUCCESS in 1h 30m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2123d2bd606144b993e87b0fc8d00c83 : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":4},{"id":"30b593cca19f670a750cb1c2f316683854650c83","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-16 07:01:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"bc1b7fd9e1b23695ab35a2bc83440c9f148079a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-16 15:54:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5807ebd5e44081074368031ec271f6eadcc0ef34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 15:54:55.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1719/","accounts_in_message":[],"_revision_number":5},{"id":"4ce34ff2c102687917d8c62ffe05c22630fc0870","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 15:54:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"58ec8ae5cc8756304a66d4b87d87b2c5e312d495","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-16 15:55:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"902c48ef409f0910f6dd42eaba009a6a9bfd6a49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 16:12:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f0ad47044f4695b3b58760e86befca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bab2ef89c2e040da8cc6209420ca082c : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19446a3f99214e1192644063119a4df8 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/16fd34340bd94a038de9a9af78f1c599 : SUCCESS in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"30204bbef5884d9fb6a6e00a5db6df29d58fc19c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-16 16:41:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946995 : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":5},{"id":"9ffab40374f367cab8474b494feb1d5057c20ba0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-16 16:55:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946995 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":5},{"id":"be0ed41e78196d36545257813bf0f955b5111fec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-16 17:34:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a2d66f6836144f4e8456c9a7caeec444 : SUCCESS in 1h 17m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ea57a0dbf4e04d028eb3dcb591f7043b : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":5},{"id":"4971df9161c737ec8eca8f95e233dc3248dcf25d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-16 17:39:44.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c8b9c96febe14a35bc668aed32b172c7\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7434cc0c128443f8b6938f676000d33b : FAILURE in 55m 17s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec9d899e0f2b48258a44eaae2405bff8 : SUCCESS in 42m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bccc1142000c45f88a07cdad8184acd8 : FAILURE in 1h 44m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c72620a583e14f6cbe32f338a0d2557f : SUCCESS in 42m 25s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8dcfeee12fc444849b280e1f94449294 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b49416eac314ae583ffd860df74d31a : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16445c2b0c664803a60223b7e72a78b4 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ad9cce385e94389a30c775739951b88 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-d... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"c7a3fc9b7414c086ccd747a5b32013a29687f4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 18:07:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24d3dfb2392245cf95accd4f6ce24ac2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6bfa46ca9f14782a0b42b80a0511e80 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a208cd0da70e4955b54b46063330f0da : SUCCESS in 8m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53ad825bc7a54322b729c10eb705ada1 : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50fa7e7369da497e91b9cb6cdecd575f : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8f30abeffc43818e047bc93fda4aa9 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/04811cd30a5443f4b17999f7a0fdc6b2 : SUCCESS in 50m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eed6fa11f5d34eb589925414a6533123 : SUCCESS in 44m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d059a3baccb0411e94f32ccffc7092ba : SUCCESS in 1h 25m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c95a081cfeca4a14a1d356ccf9c6252d : SUCCESS in 47m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/71a45dd3f40646afac90be593e8fb9a2 : SUCCESS in 26m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8873331b51d46478c2d0f913b43deab : SUCCESS in 7m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c1322acd48f436f84abe69fa5646b95 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f812b7cb0f54237a3242ea475ce7883 : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/66bd9eadc7154a2d84ee729bdf91df09 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1d54ad4f224042aa8116759715041595 : FAILURE in 18m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fc1f342baef48799d1b4590183550dd : SUCCESS in 7m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8fed02e85444ca09a663dbc853cb175 : SUCCESS in 1h 55m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f037207b26548d5a4bb4e171d4749b7 : FAILURE in 1h 26m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a9685dc69e9420ea86b6f46e86b0a3f : SUCCESS in 1h 31m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/253e68e052a74281bfca2a1f0099c5f9 : SUCCESS in 1h 45m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8c95d59d2de44340bed6bc2c66fd1efc : SUCCESS in 39m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/381e894317fa4d3e9934426a7f920e22 : SUCCESS in 1h 08m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2b573db0caa4f46a482abcfa565c5a1 : SUCCESS in 1h 28m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b47995e0a904f818d687067356f0374 : SUCCESS in 37m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ead0a9d2617e4fcd9696535f4c905e4d : SUCCESS in 1h 31m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e07587e9480c462fbb63978f97f487dd : SUCCESS in 54m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5844ebdd04547859388fb0e94a661dd : SUCCESS in 2h 06m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f93d75eeac845e7a5d3a951dbfb4f28 : SUCCESS in 1h 23m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58554e6c6a6548fc95759ec294b12e7b : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b21e88ce46cb3092da46541ac33b07381ebdd39a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-16 18:08:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9002abef14133671acab8efc0a3a56de4620b04b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-16 18:50:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/4/20250416232318/NEC-V-FC : SUCCESS in 3h 59m 35s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/4/20250416232318/NEC-V-ISCSI : SUCCESS in 4h 26m 53s","accounts_in_message":[],"_revision_number":4},{"id":"25c4af78653714b925f2e03dee8568ac029d6793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-16 21:17:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f363fd29e0b0cce1adc71efe22fbc652d73eb6be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 21:17:51.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1720/","accounts_in_message":[],"_revision_number":6},{"id":"7c7f86ec6043f8da28afecbc6f47248e91c7862b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-16 21:17:53.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"4d39308f3004c4661e2993420265e605c604d6b7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-16 21:18:33.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"df145412a3d3d955f60eb6b1c8e7ca3d5652b6b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 21:34:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8831ad9903e843e0ad46547762f40dcd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ca923c8c41d4e0aa9cbf866db58e333 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0de369e427344dce9488ff8d8605d604 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6874f8915c548db83dad3456f57ae8d : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"34d2e6d5dfa220cd72670c97b7cfcd9d7a53cc3c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-16 22:06:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946995 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":6},{"id":"7217491fc26b41895d536c6a6d22b9019615fb1d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-16 22:18:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946995 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":6},{"id":"0ff750505119ea61e582c91b47f7a8dea9ddb669","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-16 22:31:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f0312cbee65a4cf08971986bfe64d8d7 : SUCCESS in 1h 09m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cdbd72e5eb074d31b3b45c63082b8cb9 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":6},{"id":"cf0b3aef93fb3f6058f1cf376e2fddc130390cde","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-16 23:02:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fa6380e1fcb64e27ad6f18737ed0a64b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f266860f6efc4d3e93931a2882343409 : SUCCESS in 39m 57s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62eed9cdbf354fafb5f50c4612a55c98 : SUCCESS in 52m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/363af06201ec46abbf65dac8b62072ba : FAILURE in 1h 44m 11s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9ce8f85a97b4c94a56b64a171fcdb75 : SUCCESS in 43m 18s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/56542e74a50d46999af3c26419a63a7e : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2bafb962dc1943f6a9e3aff1f13949d7 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/50416ccd13e546c5b348468f22f53054 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b07baaa8d5744e3b4c4be894ae1ae82 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"46183fbd3f3fce8070d95e6680db2aff385922e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 23:18:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00970ba14ecc43889826f11b13d5ae3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e4a50795bd4d2d9696ec435ce1b8c4 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/843c3910f2164ae288993b84834d89ad : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d179708d507d426f84d940b8183fecf3 : SUCCESS in 7m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3aaf4d347a6948b89e4788d336a540cf : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685fb6cb03a34676a5a00ad3bd52d5ca : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1f0e74c5427b4be983f15a857afe8116 : SUCCESS in 53m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e67508b03c3484e8e2e08c9a97f117c : SUCCESS in 57m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c53929e75de42beb6fd7a61c3c08d3d : FAILURE in 1h 28m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b419e32be8714242b4d5418e29ed7e37 : SUCCESS in 1h 00m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f6083140254d46f98fa2b683701eaa67 : SUCCESS in 25m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cb58c420123f4d57b40f6dd4c33ae965 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5906ee583db47feb015f175c2da30db : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b695bac57c6a4935bab3fab1609a1005 : SUCCESS in 5m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0bcccb89dcd644cd8bbb466ab033431c : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67b29f014f5048958530dd27764a7f10 : FAILURE in 34m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a7352c554ae414ba16d8dd170fd5330 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fbb50c38f3e34e3580f7c72fc7d575b4 : SUCCESS in 1h 29m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01be7e2eff794efba5e39de19c82965f : FAILURE in 1h 57m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15c87ac85c45407bb8fbfa10c66e37ec : SUCCESS in 1h 55m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7fdd0f902e6a475dbe4c3528c002d885 : SUCCESS in 1h 55m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74050a380c1e4755bf3e0defbdea7d40 : SUCCESS in 22m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f0f74fce32249fb8c9c4e880f76c3c6 : SUCCESS in 1h 04m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b44e92419374ffb9832b8ced31e2134 : SUCCESS in 1h 39m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3ee7a708c30c4cea80a749181137f607 : SUCCESS in 44m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3d3097e579f44a4b723125bf3a843bc : SUCCESS in 1h 27m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5f2d11b727f24d5793b3554f74b13ae9 : SUCCESS in 1h 05m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bcc0e0583df44ffea2053b5e05e76b98 : SUCCESS in 1h 54m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/783aae21fff5467c9a83975df7a3992a : SUCCESS in 1h 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/389877b44814491ab6731fee34993c17 : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":6},{"id":"39b44cc4fb5d7ac495f6f8fdb623f3dc1c24992a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-16 23:57:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e1957f656f73ef02ffa748d823571e65d42a5bf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 00:13:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fdf308773864a39886036d2d150f9de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b09777f380140c79256d777db2d0ce6 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0d735d5756e4f148edc831e0636b662 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9243a7b466dc4a388ef4e1f9c547bf0c : SUCCESS in 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5e7b2f2c152ac8086015e59f161e9681f3692290","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-17 01:30:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cff3f1477d4b46bc8a2013766f7c35e5 : SUCCESS in 1h 20m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9afa1c7657474471a70c859ca643df91 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":6},{"id":"9435ddeb1d83eeb0fd86c2d0fbf238a4b055e6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 01:45:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f31cc3a189d9467996ce4befd84ceb78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07467d46e9a04d76955fc45e5c4f3ecc : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5caedf76b0ef432aa62e6b31aa724701 : SUCCESS in 14m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7ef972e32c94a958db31b24c3ec76d3 : SUCCESS in 11m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab777a9189b04fa787104e7fbc078fcd : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe415fc5e9974e2881248f1c382f720a : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/6b3a4217c95146a194ac011839097a00 : SUCCESS in 56m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f6ab314430e4e3489be3a3f62266519 : SUCCESS in 56m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d0c3fe0d8f043a984de2ebc98c7300b : SUCCESS in 1h 19m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ada7701768964641a638fe27835361e7 : SUCCESS in 1h 00m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a38fbc515caf4cdea44050bfc12048f0 : SUCCESS in 22m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5aaf2bfc7ce495b810fa49a48f7f93f : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11a857cb705c482c83b3548f59fafdbb : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81c32b5b2340431b929dfffc45d0731a : SUCCESS in 6m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1bedbb3ec91b40b1888be95457879868 : SUCCESS in 3m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a3f82c268df4a48b62570ca48010263 : FAILURE in 28m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e11bcf8410114757944855fc01b6bcd6 : SUCCESS in 3m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/90fca0dd13b549ed8daed14c638c8865 : SUCCESS in 1h 37m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/986c722f6beb4709a2620388fe5b6c98 : SUCCESS in 1h 18m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e90e0c711d794683ad6ffffca7317cf6 : SUCCESS in 1h 27m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/44b136a8aa164d35a6a1b854e9558224 : SUCCESS in 1h 29m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2b66f8ac37ec484fae799d444ff98f34 : SUCCESS in 19m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7c29faf5d1ac484897d54117a58e6f8d : SUCCESS in 48m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8541cef9331747038ccc98b2c79ab135 : SUCCESS in 1h 31m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/07a6babb6fc24be9991d64cbe1e450b4 : SUCCESS in 32m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2be74db540cc498db5d205fa37cff944 : SUCCESS in 1h 16m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4b77b56270594b448a2e1827c6f46243 : POST_FAILURE in 1h 34m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0723907356a045b5ac840aebe6f93e9b : SUCCESS in 47m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e02c64a1651642359575f4b5b225497d : SUCCESS in 1h 10m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcc6b62f4a3444728ed15f9b4585fac0 : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":6},{"id":"455da5184c95f60db85ae15873af070510641988","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-17 04:46:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"7b3d37385dc2c2129d4939deba0b970ef1a5fef1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-17 05:36:16.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-6/FC/ : FAILURE in 8h 18m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-6/iSCSI/ : UNSTABLE in 2h 3m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"81dfe0d9e62a703bdd9dfc998b2bfeaab624f50c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-17 06:56:32.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-5/FC/ : UNSTABLE in 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-946995-5/iSCSI/ : FAILURE in 15h 1m 13s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"c00661c29197322f3d0205cfd779bb7828aee678","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-17 10:12:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0794e80ab13e651b96b5fa39318c297a8d0a0a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-17 12:03:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3de999a59c093ec4522506913ee5cf7f1b1da3f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 12:03:32.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1722/","accounts_in_message":[],"_revision_number":7},{"id":"967afc9aaa39084bffb956cfe2913c20defb9cd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-17 12:03:35.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/946995/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"da8722507d14f58ea01350b43cdcab4b6b62bf11","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-17 12:04:13.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"265e9a26e63e0304b8c44232288aefd7119d1b66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 12:23:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4f32c4d790496cab4e5a6c49aad7d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c87aa02d76434475b935b17d1f0b143d : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ce6dac3e1d846c1bf53fb1c673eb555 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b6c427e524014717afbf8b3fd455823a : SUCCESS in 18m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0896eb8ea8069fb8c6e1b6790a31793bd0b102d8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-04-17 12:33:04.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4c47c27297e5dadb6ba833965e432fe1a07ebb08","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-17 12:49:10.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4ba83d6e915413e2f37a167474ed9dd082948f83","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-17 12:50:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/946995 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"758c819b40ca1bcd2d972f91c4a394c1a7d929f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-17 13:05:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/946995 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":7},{"id":"dc6b745e9c9eb276e13e1a713915ea6a3ed5a338","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-17 13:22:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f93f7ea0a39e4c7986addc8ced8e3548 : SUCCESS in 1h 17m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ff4615b9525840b6bb72d135f7d1f11e : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":7},{"id":"9c829304b8189f3181bbbd0c8fbb1605b7569930","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-17 13:36:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/809643c7b2b04ee3af7a9d2570b163b4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/835964899b404cdd947aea74d60f087a : SUCCESS in 39m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a4efde98649640759a6af62bead88051 : FAILURE in 1h 19m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a672823621164966bb83dfbff31c3055 : FAILURE in 1h 32m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e81e32549f944ab9df94163818fcf20 : SUCCESS in 41m 57s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/59ab956fe55d45739cba137417694405 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33130b36335b4e1eafb25efa36b46423 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8498dcd4b68d4833afbdb14efe19dab3 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90e56bef6d3e43518d2dd3cede685f9c : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob ... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"b7fe314ddc2d3d731575eb4a49e7de9738422e46","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-17 14:20:16.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/726/.","accounts_in_message":[],"_revision_number":7},{"id":"9c980bf69d273a041453eb5d8e3cf5e21096648a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:40:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf808ce737214cf08ba6240c427be947\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9b7271f9c04c2c8434e4e4477c62f1 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf7b270b8a804d00884e044e1a151b84 : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2531ca32965d4d7baf0c4d3b9d9f1c0f : SUCCESS in 12m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97966297679945b9b56a9f10032a60c4 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f6fe796edb4c71953efc779ad600a5 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/18e13b1191784a5aaae0ebe5a46a72f0 : SUCCESS in 54m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04ce91a7af7548f8af055164d5786aa1 : SUCCESS in 28m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/410498ccd17b42099f915caa59cbebe4 : SUCCESS in 1h 24m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8dcb348496a4adea5477c1b9a0261af : SUCCESS in 57m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4b60658447c471591456f5ca4ea1230 : SUCCESS in 20m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a73a4c23b823469ea01bc32006611984 : SUCCESS in 17m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8425a3cb43c4390b47ae4faf5dbfb72 : SUCCESS in 3m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a7c0ffef242492bbd947d549fc3d8dd : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/332f4f23fa2049c1b1e603c64def4c36 : SUCCESS in 4m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/311d8344d5bc4d73b4552cddfc0ca2a1 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b0a1568401f4291abbea103b03a460f : FAILURE in 31m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/622b7200c40c4b39a5213b9e09e623d5 : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0dd3c3aa6a64e3cb9c84aa52f619cd0 : SUCCESS in 1h 49m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/45833a5d5d0940d5bfe0ca8e2405c107 : SUCCESS in 2h 31m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0e35f996f0948a4a220a2583cc2a2c2 : SUCCESS in 2h 01m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/72c5d911059d410e90994d35a6331339 : SUCCESS in 1h 27m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b7e67aa751644bcb07ae276becbb345 : SUCCESS in 34m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27c1b078458b4047a1ac71e0828e8f04 : SUCCESS in 1h 02m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/941446bfd336448f8402bf23cd3bcf94 : SUCCESS in 2h 07m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a05f8660fe3445b3a820d09041c8cdae : SUCCESS in 43m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0da53a4880345e69975fc9de90d36f3 : SUCCESS in 1h 27m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a3462233a2ff499c89ba429d954f9744 : SUCCESS in 1h 54m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bf4be7d3e97434997c666e3e2d306fd : SUCCESS in 1h 45m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b81712873cec4364a1ca3622213481b4 : SUCCESS in 1h 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e36df4e89dab4043a11149fe9fc48406 : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":7},{"id":"e8bfe3dd0e4020bce189b0ea2715eb10a13ddd3f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-17 17:48:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"8a16b0f8e46c44bfdc3ce28d55db72fc2d7d27e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-18 06:32:36.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/7/20250418070621/NEC-V-ISCSI : SUCCESS in 4h 21m 02s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/7/20250418070621/NEC-V-FC : FAILURE in 8h 26m 11s","accounts_in_message":[],"_revision_number":7},{"id":"3f9479b15943b94c93dc74446fe94becae7c2d6f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-19 04:09:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/7/20250419071217/NEC-ISCSI : SUCCESS in 5h 29m 44s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/946995/7/20250419071217/NEC-FC : SUCCESS in 5h 57m 04s","accounts_in_message":[],"_revision_number":7},{"id":"9f4ce513b2cbbc80ee28c232c3c23ffaf51f48ca","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-21 14:00:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/946995-4/HPMSA-iSCSI/ : SUCCESS in 2:19:54\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/946995-4/HPMSA-FC/ : SUCCESS in 2:06:45","accounts_in_message":[],"_revision_number":4},{"id":"4bfe09e9a3e25542339953ad5e62337fc3389e0f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-21 15:30:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c756bf6b812236d06f0ee2d1f2f59f9c084350ba","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-21 15:31:13.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"562517790656479b0ce6d610146b454da86393c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 15:31:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d14408b00d6001cbfcec0c019731336f8a374a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 17:08:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9e9c86375d44578240aed4bf80c2a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a31b3626d34b638af43472cc0f7766 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5b57a3b197d45f8b0f7b734a20a4103 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a9b99e26bf84695929abd322a12b443 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845af7b59ea94a6994f05d778cb85b84 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0fb34dbe6de2484e8a4deb2d0c785f6c : SUCCESS in 29m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7872908a1a52426994b948f32c340f68 : SUCCESS in 49m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ea48dedb5134328b2a4e2636d9a2305 : SUCCESS in 1h 17m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a8aad27cc3b445d9ab1efbfbc4f6205 : SUCCESS in 51m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dd0a5df959e471cba9d9c69bbade72e : SUCCESS in 15m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8f51f6aaf1944ceae4cb5046c5fb83a : SUCCESS in 31m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dee5452e58c544afb1d1a5b433cff6f1 : SUCCESS in 1h 36m 46s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbfc297b46d74b8c8c18f87ca01e931d : SUCCESS in 1h 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdcbc122051b4e57bc8edf9c85c7534f : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":7},{"id":"f9a88609767e61075988449cfe3f0cb936d47f00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 17:08:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"f7e51a9497cae487e5eaab6d357c61a9fe62690f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 17:10:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a1e9721b28b462882de39072f0181b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc2457a59d649c0a3b696f3722c2747 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91c20c2cee874203b19f91e1f44274f7 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":7},{"id":"bb1cab07a3d813bf047d5998c6332368ff40c0c4","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-04-21 17:25:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/264c789b8fc444fc9cea462a57ac969c\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/2609bdf1c2c947edb775f269a06b21a2 : SUCCESS in 1h 54m 09s","accounts_in_message":[],"_revision_number":7},{"id":"ef03064aee351807ec5e425a91050a97bc005398","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-21 18:48:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/946995-4/Lenovo-iSCSI/ : SUCCESS in 2:20:14\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/946995-4/Lenovo-FC/ : SUCCESS in 2:07:16","accounts_in_message":[],"_revision_number":4},{"id":"a6dd02e817c205e86cd43c23451916dda55ba8b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-23 12:40:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0aeaec83c04c1231d6e1717605c9624b4bd3335b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-23 19:19:35.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"d6a6f3b74e8a5bd96a0951269ab0753c89936c5b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-23 21:23:51.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9687fbac79b0af266dcefb89b8ecc5c2940f6c80","revisions":{"d01ba874929c9631d1457744ea34bf63db9bfb9b":{"kind":"REWORK","_number":1,"created":"2025-04-11 13:07:19.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/95/946995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/1"}}},"commit":{"parents":[{"commit":"dd95e1ad7da275692198fdbb20536021e056ef52","subject":"[docs] update description of mv 3.46","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd95e1ad7da275692198fdbb20536021e056ef52"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"subject":"NFS driver: Fix fail creating volume with multiple snapshots","message":"NFS driver: Fix fail creating volume with multiple snapshots\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d01ba874929c9631d1457744ea34bf63db9bfb9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d01ba874929c9631d1457744ea34bf63db9bfb9b"}]},"branch":"refs/heads/master"},"e28ac87595361e49fcdeeb86bd5f7dd81cb16c79":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-11 13:12:31.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/95/946995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/2"}}},"commit":{"parents":[{"commit":"dd95e1ad7da275692198fdbb20536021e056ef52","subject":"[docs] update description of mv 3.46","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd95e1ad7da275692198fdbb20536021e056ef52"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:12:31.000000000","tz":0},"subject":"[WIP] NFS driver: Fix fail creating volume with multiple snapshots","message":"[WIP] NFS driver: Fix fail creating volume with multiple snapshots\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28ac87595361e49fcdeeb86bd5f7dd81cb16c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e28ac87595361e49fcdeeb86bd5f7dd81cb16c79"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c9d3b2319cc66b26337b07acc7824aa5da0012e1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-11 13:19:27.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/946995/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/3"}}},"commit":{"parents":[{"commit":"dd95e1ad7da275692198fdbb20536021e056ef52","subject":"[docs] update description of mv 3.46","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd95e1ad7da275692198fdbb20536021e056ef52"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-11 13:19:27.000000000","tz":0},"subject":"[WIP] NFS driver: Fix fail creating volume with multiple snapshots","message":"[WIP] NFS driver: Fix fail creating volume with multiple snapshots\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\nCloses-bug: #2074377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9d3b2319cc66b26337b07acc7824aa5da0012e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9d3b2319cc66b26337b07acc7824aa5da0012e1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"541b983355c486ad64defe03373fc26887bd0420":{"kind":"REWORK","_number":4,"created":"2025-04-15 17:00:26.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/946995/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/4"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-15 16:52:23.000000000","tz":-240},"subject":"NFS driver: Fix fail creating volume with multiple snapshots","message":"NFS driver: Fix fail creating volume with multiple snapshots\n\nThe NFS driver uses qcow2 images with backing files to represent\nvolume snapshots, which is not allowed for qcow2 disk images\ndownloaded from glance.  The driver uses cinder.image_utils to\nconvert a qcow2 snapshot to a raw volume; this was not a problem\nfor the first snapshot, whose backing file is raw, and hence passed\nthe image format inspector, but the second snapshot has a qcow2\nbacking file, which the image_utils were rejecting as a security\nrisk.  Thus we now pass the qemu_img_info from the backing image as\nan additional parameter to the image convert call, which indicates\nthat the file has already been screened and allows the conversion\nto occur.\n\nCo-authored-by: Fernando Ferraz \u003cfesilva@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\nCloses-bug: #2074377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/541b983355c486ad64defe03373fc26887bd0420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/541b983355c486ad64defe03373fc26887bd0420"}]},"branch":"refs/heads/master"},"33512d087937bcbc63735a68572e9e05e744cf70":{"kind":"REWORK","_number":5,"created":"2025-04-16 15:54:48.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/946995/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/5"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-16 15:54:39.000000000","tz":-240},"subject":"NFS driver: Fix fail creating volume with multiple snapshots","message":"NFS driver: Fix fail creating volume with multiple snapshots\n\nThe NFS driver uses qcow2 images with backing files to represent\nvolume snapshots, which is not allowed for qcow2 disk images\ndownloaded from glance.  The driver uses cinder.image_utils to\nconvert a qcow2 snapshot to a raw volume; this was not a problem\nfor the first snapshot, whose backing file is raw, and hence passed\nthe image format inspector, but the second snapshot has a qcow2\nbacking file, which the image_utils were rejecting as a security\nrisk.  Thus we now pass the qemu_img_info from the backing image as\nan additional parameter to the image convert call, which indicates\nthat the file has already been screened and allows the conversion\nto occur.\n\nCo-authored-by: Fernando Ferraz \u003cfesilva@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\nCloses-bug: #2074377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33512d087937bcbc63735a68572e9e05e744cf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33512d087937bcbc63735a68572e9e05e744cf70"}]},"branch":"refs/heads/master"},"9ddd4c3c8fb8ac169643b47eee34559b1e79c71a":{"kind":"REWORK","_number":6,"created":"2025-04-16 21:17:43.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/946995/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/6"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-16 21:12:20.000000000","tz":-240},"subject":"NFS driver: Fix fail creating volume with multiple snapshots","message":"NFS driver: Fix fail creating volume with multiple snapshots\n\nThe NFS driver uses qcow2 images with backing files to represent\nvolume snapshots, which is not allowed for qcow2 disk images\ndownloaded from glance.  The driver uses cinder.image_utils to\nconvert a qcow2 snapshot to a raw volume; this was not a problem\nfor the first snapshot, whose backing file is raw, and hence passed\nthe image format inspector, but the second snapshot has a qcow2\nbacking file, which the image_utils were rejecting as a security\nrisk.  Thus we now pass the qemu_img_info from the backing image as\nan additional parameter to the image convert call, which indicates\nthat the file has already been screened and allows the conversion\nto occur.\n\nCo-authored-by: Fernando Ferraz \u003cfesilva@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\nCloses-bug: #2074377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ddd4c3c8fb8ac169643b47eee34559b1e79c71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ddd4c3c8fb8ac169643b47eee34559b1e79c71a"}]},"branch":"refs/heads/master"},"9687fbac79b0af266dcefb89b8ecc5c2940f6c80":{"kind":"REWORK","_number":7,"created":"2025-04-17 12:03:22.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/946995/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/946995/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/946995/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/946995/7"}}},"commit":{"parents":[{"commit":"db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396","subject":"Merge \"trivial: Remove trailing spaces, mixed tabs/spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db6d00e8cc4a8de2b03eb35fc0864f6a9b7d4396"}]}],"author":{"name":"Fernando Ferraz","email":"fesilva@redhat.com","date":"2025-04-11 13:06:26.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-04-17 12:02:27.000000000","tz":-240},"subject":"NFS driver: Fix fail creating volume with multiple snapshots","message":"NFS driver: Fix fail creating volume with multiple snapshots\n\nThe NFS driver uses qcow2 images with backing files to represent\nvolume snapshots, which is not allowed for qcow2 disk images\ndownloaded from glance.  The driver uses cinder.image_utils to\nconvert a qcow2 snapshot to a raw volume; this was not a problem\nfor the first snapshot, whose backing file is raw, and hence passed\nthe image format inspector, but the second snapshot has a qcow2\nbacking file, which the image_utils were rejecting as a security\nrisk.  Thus we now pass the qemu_img_info from the backing image as\nan additional parameter to the image convert call, which indicates\nthat the file has already been screened and allows the conversion\nto occur.\n\nCo-authored-by: Fernando Ferraz \u003cfesilva@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nCloses-bug: #2074377\nChange-Id: I49404e87eb0c77b4ed92918404f86c073fbfd713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9687fbac79b0af266dcefb89b8ecc5c2940f6c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9687fbac79b0af266dcefb89b8ecc5c2940f6c80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
