)]}'
{"id":"openstack%2Fcinder~947663","triplet_id":"openstack%2Fcinder~master~Ia486448b85feba636effef609d79ef8e57a6d39a","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-29 11:51:10.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-08-29 18:07:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2025-08-29 04:22:35.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-08-28 19:18:19.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2025-08-29 19:49:47.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-29 15:08:54.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ia486448b85feba636effef609d79ef8e57a6d39a","subject":"[NetApp]:Cinder support for self-signed transport","status":"MERGED","created":"2025-04-18 12:24:41.000000000","updated":"2025-08-29 20:04:17.000000000","submitted":"2025-08-29 19:49:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947663","meta_rev_id":"e6ae39673feebad8ad9bf30337b3897ff9fefc00","_number":947663,"virtual_id_number":947663,"owner":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 19:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-08-29 11:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-08-29 18:07:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-08-29 04:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":1,"date":"2025-08-28 19:18:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-29 18:06:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-08-29 18:07:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-29 18:06:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-08-29 11:51:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-18 12:24:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-04-18 12:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-18 13:07:23.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-04-18 13:11:36.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-18 13:26:00.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-18 13:59:50.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-18 14:12:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-18 14:13:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-04-18 14:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-18 16:50:47.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-04-18 20:01:54.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-04-19 16:00:50.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-21 20:33:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-21 23:18:32.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-04-22 18:14:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-04-24 00:54:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-25 17:24:14.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-26 01:36:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-26 09:48:41.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-08-10 14:39:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-08-10 14:39:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-08-10 14:39:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-10 14:39:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-08-10 14:39:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-08-10 14:47:38.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-08-10 14:47:38.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-08-10 14:47:38.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-08-10 17:45:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-18 06:03:11.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2025-08-19 04:05:48.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"REVIEWER"},{"updated":"2025-08-29 18:06:19.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"d80913f1af1b6b4106f49b77017cd0b6a59be8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-04-18 12:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff710db74b90e271672b5871717a5b657523a9eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-18 12:24:50.000000000","message":"Patch Set 1:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1725/","accounts_in_message":[],"_revision_number":1},{"id":"89da644558562435958a8ee8d1c72c2558e1187e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-18 12:24:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"e518dd0b137e439a4904e9640e326a3f8c1606b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 12:43:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c199d79ce8564fbf81f5efe77ca97c9a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc2863c9e4b943939928c7dff2ebe97a : FAILURE in 14m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a7c2620407e4d84ac662f4d72f5667d : FAILURE in 13m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7a90efb74954c28814cb7ff14731dff : FAILURE in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"747df0f13f15cfe1d6c39591a5bc2dbf2bbb6c11","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-18 13:07:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/947663 : FAILURE in 39m 40s","accounts_in_message":[],"_revision_number":1},{"id":"052855b29f01b859c1b5908ac4bc13006db31aa5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-18 13:11:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947663 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"053e1960b9a5468f43e4e0a778b4b0b5f43ccb06","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-18 13:26:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e10651613b81050d93262f3885fcf04bdc0c1c8b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-18 13:59:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/39dc807ecc234aaa9353abea7c4bd9ff : SUCCESS in 1h 10m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a345015ac5224b228cc60805eb453850 : FAILURE in 1h 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"25d1a8e80e96dedeffa1be2793c8ee6222ef876f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-18 14:12:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8753a2aa36074eceb25a49e951996e0c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/265c81befff34fbbaab520dcc764cf7f : SUCCESS in 44m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ce8a500eb564d60adcba42d1219365a : FAILURE in 1h 06m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d1840e6cd5e94096a61b11b344dfc255 : FAILURE in 1h 47m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8eb9487ba60948e1adf133333d8c6086 : SUCCESS in 38m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a67a49e93f5a4e5994ac4f56ed490f7b : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c634bac690b04b7fb4d3807f984798c3 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3b02bdf51944450958dc2909a83ac70 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1e119a8af6e64e17ad5ae67c6fb61ac5 : RETRY_LIMIT in 25s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob ... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"1d5ae89eef0a3894ac98419c454397a5467f008b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-18 14:13:06.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/615/.","accounts_in_message":[],"_revision_number":1},{"id":"a52ba75c97f9506b4ef1b3ea7116ed4b2291b7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 14:41:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52a651b8ec8043749b4b63b1113c52e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fb829215c444409a18e0c6f232af14 : FAILURE in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05860fdf5bec485e9436de12bdec036c : FAILURE in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cd148c7f96740138e4ae5b7d43f9476 : FAILURE in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b005f0070ba94f2b9111273ddffa9fd9 : FAILURE in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6884f37a077454286dae1763fb74d53 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/438a2d2ca5134a2789fc6042d6d2f94d : SUCCESS in 47m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29c847eb345941faacf77260fff084b5 : SUCCESS in 30m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33af134320724fd994a9d9a32a9958ab : SUCCESS in 1h 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8401d750aaf7429fa3fed14be8896154 : SUCCESS in 31m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd7049b639044a8e9b7414b042e22f41 : FAILURE in 23m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c405cf26e906445a966bac1e414e4e0a : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/55468a99007749cebb9329ba548bed81 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9f0946caa5945049fd5fcf622ad4877 : SUCCESS in 5m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/127828a66dfe46fb8ff67ca4dfa60116 : SUCCESS in 3m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6394db07965d43b3a3b286606ea5b575 : FAILURE in 32m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08cfd5b13b5244388277bcbd87de7bc8 : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab29296a808641d6ae139e8b145afe21 : SUCCESS in 1h 44m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b22f3427e554319a7836655468804e5 : FAILURE in 2h 14m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/acaaba8c7b834555a49802fe3555db4c : SUCCESS in 1h 29m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/26cd7fe1aafa4c70980028d1f005739e : SUCCESS in 1h 42m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dcd12269626a430bb8ccfc5c0942674c : SUCCESS in 33m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e9fcd19942c491aba50bb167dffaa44 : SUCCESS in 52m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a6d2a98c350a4c4bb9a44c981c8b2e68 : SUCCESS in 1h 18m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7dc94a4c30fa49b9a350b48d5762c7b9 : SUCCESS in 37m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3fe135e391a645288ba0cd71e996b901 : POST_FAILURE in 1h 53m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bb20a6a354774959a325464800803903 : SUCCESS in 1h 19m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a9c11074d7e499eb9f1d01bbdde26e1 : SUCCESS in 1h 14m 31s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1c5dc8f81284ce9a614f05125ae9e5f : SUCCESS in 1h 28m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de7f6884b5c848cf9243acaa01c8e29e : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1d8d76d7f70451441f526db934f86f3105bb42b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-18 16:10:49.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/736/.","accounts_in_message":[],"_revision_number":1},{"id":"eb8937e5681f8da766399520a654440144a8effd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-18 16:50:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/1/20250418212444/NEC-V-FC : SUCCESS in 4h 13m 01s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/1/20250418212444/NEC-V-ISCSI : SUCCESS in 4h 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d498b1e95c5c0e5362e6bf17326645f3270e9891","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-18 18:03:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/617/.","accounts_in_message":[],"_revision_number":1},{"id":"e6308a95d358d529b69096b1bf04db22e14d71f7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-18 20:01:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-1/FC/ : FAILURE in 7h 36m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-1/iSCSI/ : SUCCESS in 1h 28m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"a91d5614113da29c5a9ccb780f2a31b7ef885206","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-19 16:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/1/20250419190421/NEC-ISCSI : SUCCESS in 5h 27m 42s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/1/20250419190421/NEC-FC : SUCCESS in 5h 56m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0c42a7af941943127ad8c42eea3339a2818821db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-04-21 15:22:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eeb9044a63c63d95fe5e5562a2f22900b28571bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-21 15:22:54.000000000","message":"Patch Set 2:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1728/","accounts_in_message":[],"_revision_number":2},{"id":"00bb29a06b37a8738b200c93a152888c055c6be2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-04-21 15:22:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"8da0f99c4ed8301ae48bb261cff395943fddab6b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-21 15:35:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/947663 : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3877ec3e99086f7995684cc7de4ddbafad979262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 15:40:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d33cdbf3904abcb0219b777f8c50ab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea9f7f2e555b42918d4297f5c314bfa0 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95d62ad983ef494aa4271f99246e65a4 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8303d02380e54e3daca7b1a86fd5016c : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f64d73c2b429158155e4214ae3e3ecc8fffca28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-21 16:25:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":2},{"id":"deaf8cf2c29f27ad42a81c0159a8417fd61a81e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-21 16:46:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c314b5aaa000424591216b75913061d4 : SUCCESS in 1h 11m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/157c51501bdd45b09d1b41d2840df8c1 : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"44b46dac27145452e319a3d61d4f474dc01afc1b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-21 17:15:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/361dff6651684701a905868be3d55b82\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b853524f93784cadb8deb5172b985f8f : SUCCESS in 41m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee9ef25f4e2c42eebb9da93f58ceb333 : SUCCESS in 43m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b7847d10fa44433ade80530fe57c4f4 : FAILURE in 1h 51m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/06dda4b1d9484ca6897818ec4466e886 : SUCCESS in 41m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/437ecf1ba3464b91acd2d96e94958345 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ea72f7d572244608cc2231695f39837 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df423ff49ea9413a92df3477b7b0799a : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/808d6d00646b46cb9f9c1b4cd6ab7114 : RETRY_LIMIT in 24s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"e5311c0803b83248704dae44b774ed64c938b453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 17:31:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e7d7a64a304af4afd59e9f06b613d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6e4913008d421c9f678b911da17ccf : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9094001223494ee094a0e5e51e907b38 : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf16fde897b54edd880b8a08f0d32bc4 : SUCCESS in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10dc5d1005714316bb2077b5abdad93a : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05789083a1594998be5a0c617e85dbd5 : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/49e934002fa74282a43e51713513ed57 : SUCCESS in 29m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/539a8f39c68a46ffb514ad84f6df87de : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3197e8e5dfb84eadb31869eb3d218aeb : SUCCESS in 1h 21m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ba19974a2724de4860c9fddeb5eb2a7 : SUCCESS in 31m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d41555fb9dbe4f3d84b982a5c20ed999 : SUCCESS in 21m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/706473c136ea4c408d92219b899593d3 : SUCCESS in 8m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/702a0d5eadba41509a9b68771fb2238b : FAILURE in 6m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6822954d9cca47358228193700a52552 : SUCCESS in 8m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/841c55fff24a4b228ebc4c0d18f0ef52 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9474cd37124476f8e12b5ff8093fb84 : FAILURE in 21m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5d69a02a1414c50b55d27c1884f4c19 : SUCCESS in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cf2e500cd74a4f7d873c01f42ed29cc1 : SUCCESS in 1h 42m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e3f3a82178ef4f8889b8694e5bcd7f39 : FAILURE in 2h 07m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17d26f6d54834b8c9ab93dbdd016377b : SUCCESS in 1h 43m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e15a51867a104a8291c870b6893b79e0 : SUCCESS in 2h 03m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf2929c370e5424fa495a22656e3295e : SUCCESS in 24m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb9a152919b24ba7b0e00d35b7e5ac53 : SUCCESS in 48m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1285641520c14df8b483c06d3518df18 : SUCCESS in 1h 38m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6348324c481749b69aabf27883194f68 : SUCCESS in 22m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0db65ba66a914a60a32cc0e6e0811a56 : SUCCESS in 1h 39m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4a8ea0250f504636bdf0cf08675c04eb : SUCCESS in 57m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fca1b498374413bb0a44d058fcab7ed : SUCCESS in 2h 02m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5f8780588bb4f57962b51ac4442a9dc : SUCCESS in 1h 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22e4311dd799404c9e52d0474062080d : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4b8c68f45ada301392912d4c4def41467c8258e1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-21 19:38:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/2/20250422002246/NEC-V-ISCSI : SUCCESS in 3h 50m 15s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/2/20250422002246/NEC-V-FC : SUCCESS in 4h 15m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f49e9117bc70897052233ba81162fb43b9f5d7b4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-21 20:33:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e29e50e66e2ed66fa60a9b2ddfd4c7ac82521128","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-21 21:35:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/2/20250422002248/NEC-ISCSI : SUCCESS in 5h 34m 38s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/2/20250422002248/NEC-FC : SUCCESS in 6h 12m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b7a71bb1ee0260b80d7a752afdf65739341ab107","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-21 23:18:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/947663-2/Seagate-iSCSI/ : SUCCESS in 2:00:56\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/947663-2/Seagate-FC/ : SUCCESS in 2:07:15","accounts_in_message":[],"_revision_number":2},{"id":"41a4244fbf9e64e5609c34214b032d0c39e880e2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-21 23:37:42.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-2/FC/ : FAILURE in 8h 14m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-2/iSCSI/ : UNSTABLE in 2h 4m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"118663caee22ee278acef75ba88e4305989db3a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-04-22 15:52:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a5165835513a24032c9f30c80ef9878d7ea38e7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-04-22 16:36:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/947663 : FAILURE in 41m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b267fc0d733f2531b8d4665c4a6d3289a9481140","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:36:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b26a12bf7843b3a499d76d4ac5cbb4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b0a668f55ca4a82992f3e043f9acd64 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff8f6763820c42b3b87b24466ddb3301 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fda742c624d249efbdcfee600aa6dc54 : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a9d87aef85f4a794007e1364aa02a8ad1f71786","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-22 16:40:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947663 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":3},{"id":"cf6b12a275c0b05919c475ac1a77b52e731244cb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 16:47:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/628/.","accounts_in_message":[],"_revision_number":3},{"id":"a5f1f8d57e768d76277c150896f58ea18c34a6d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-22 16:48:28.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/747/.","accounts_in_message":[],"_revision_number":3},{"id":"437a0a061b59945fb16ff04f2e507fba40a5b937","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-22 16:50:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a3398f0f2c794c1b9178442c3ef41b4e : RETRY_LIMIT in 14m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3307ff1b49fe40f4bd5630fba4a2bdf2 : RETRY_LIMIT in 15m 42s","accounts_in_message":[],"_revision_number":3},{"id":"dd9c2542886aeb7b9c6b6e449eb2fad425d961f8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-22 16:55:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":3},{"id":"74224fa5285e8db79a15a5d337a473c685cc76cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-22 17:29:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/80a734114db141b5805d32ee9e200ba9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c11740d47e394467bb4c9dc10543195d : SUCCESS in 52m 49s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e395f4d8d24e4ae8ba08d570dc655141 : FAILURE in 52m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c042573718f146c98268da622efcb2bc : FAILURE in 1h 36m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0e2a1b2b53bb4324a949e86ac1449920 : SUCCESS in 38m 29s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf9e75bfdd5a49c4a2c0adbecb0e4ea5 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7090495871954af38ca823966356de15 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e049921e6e8c47bd9f86981613ac95ba : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b11121c56474359bad68d4fa0c30b44 : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"6f8624d287160cbcb9e22f59beeb23aeceb37322","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-22 18:14:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"02a5f18f08b9a6a8428f9f874deceed0d5a43a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:31:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175c12a5bfa24685a8e2854f099896eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a73a6466804977976a2f2756174e9c : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9046a256d284921b1de00d76e0da7c4 : SUCCESS in 9m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba7f959a6f6b48bda9fff6f3744dfa4c : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f8f811589f24838a4dcf9c6197ddcce : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ae942c6710439b8b58fb9962e4bab6 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4edd4ac0db184b49935154185e66d962 : SUCCESS in 56m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81f8aa093bcc42eab985cdeb2be43259 : SUCCESS in 56m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7993bdbd51914d948178b2ab01fc1b19 : SUCCESS in 1h 20m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f67c5488e2343e895848354ac191b00 : SUCCESS in 58m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33965518b32f4138bf8f3feec0f5effb : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22d149f8b92741cb86b7e8e2a57fb494 : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e74d65fa4f74ebe91ef7fa17baf52b7 : FAILURE in 8m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ccb39d12ac64bc7a6abc0a7a617b381 : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2e336364eafb48d0b8392c410f034855 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d57c9c1b305747d0ae0170729fd95e13 : FAILURE in 37m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aaa146bf53d34c5ea7e3eefc360d10e1 : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a990d34caafa48b6ab4ab283ab55ed6d : SUCCESS in 1h 30m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3023a2b2356b4aa2b6e1b798d2e9f35d : FAILURE in 2h 34m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87b38930186746a1aa2fc934f1519497 : SUCCESS in 1h 42m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/44c0067dc0364964ab7a9682fdbe70f6 : SUCCESS in 2h 31m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/57470baf992f4dd699d9f8543611113b : SUCCESS in 34m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba09a72c52c143c381be726890b869a9 : SUCCESS in 52m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/24de773bdab744ca810ae7fe75c435b1 : SUCCESS in 1h 44m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b295e44e67444c9eb8eeeb1f4cb02105 : SUCCESS in 39m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0409232c19264de69dac676d381a0c05 : SUCCESS in 1h 15m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5591181aee3e4897967e2d776dfe2556 : SUCCESS in 1h 43m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5edafe0da217477b9fd83e6fdd8e8ec6 : SUCCESS in 1h 15m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0e280e17f18462ea5d619c06d9a00f5 : SUCCESS in 1h 23m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51b369323be9432bb19b0da260b71b9c : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":3},{"id":"eb006bdda215bc36fe46b6f95b21f71abd63bf2e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-22 21:39:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"11c711b7d50345df9219a336282db14ceb2050ff","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-22 23:41:53.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-3/FC/ : FAILURE in 7h 48m 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-3/iSCSI/ : UNSTABLE in 2h 8m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"01d421e6fe1ee32a95a8069f61efc9c2a5513788","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-23 01:18:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/3/20250423061627/NEC-V-ISCSI : SUCCESS in 3h 52m 27s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/3/20250423061627/NEC-V-FC : SUCCESS in 4h 01m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2198d77091e89a6db99b4a5ae0641ac5c69425df","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-23 15:45:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/3/20250423160900/NEC-FC : SUCCESS in 5h 56m 36s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/947663/3/20250423160900/NEC-ISCSI : SUCCESS in 8h 36m 03s","accounts_in_message":[],"_revision_number":3},{"id":"34cfc355c2288c4609b7e86ee23072fb48468341","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 00:54:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"13c253d0170b230612e763cc413b9fda4087c255","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 06:01:08.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"50acef9b176dedca69bae9351cd0545c6f6e93d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-04-24 15:35:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"e8a88947a08ecda866028a032abc2fa539f628a3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-25 17:24:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/947663-3/HPMSA-iSCSI/ : SUCCESS in 5:44:13\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/947663-3/HPMSA-FC/ : SUCCESS in 2:07:30","accounts_in_message":[],"_revision_number":3},{"id":"75c7082d56b24e0632425f5c10c65c7e15e6d351","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-26 01:36:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/947663-3/PVME-iSCSI/ : SUCCESS in 5:43:48\n- PVME-FC https://os-logs.seagate.com/dell-ci/947663-3/PVME-FC/ : SUCCESS in 2:07:17","accounts_in_message":[],"_revision_number":3},{"id":"90e41c6b69f82a678b59be791f461fd044f86219","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-26 09:48:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/947663-3/Lenovo-iSCSI/ : SUCCESS in 5:42:58\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/947663-3/Lenovo-FC/ : SUCCESS in 2:07:34","accounts_in_message":[],"_revision_number":3},{"id":"b29b55f01580283f0c83c79a28b79a49918b979f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-26 18:01:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/947663-3/Seagate-iSCSI/ : SUCCESS in 5:43:32\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/947663-3/Seagate-FC/ : SUCCESS in 2:07:11","accounts_in_message":[],"_revision_number":3},{"id":"ad72334b999cb5691fd623f0b75d94057cfac9a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-10 14:34:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b043eb453f1e2c3c4f71e3676a7e296df8f0ac0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-10 14:34:10.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/133/","accounts_in_message":[],"_revision_number":4},{"id":"285f7329648687ba385c3950145deb175526bf1b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-10 14:44:58.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"920aac548e2632f8fb10a1253d5c9ee23ae52f70","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-10 14:52:43.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e40a6e3d1ce5c563f1266d70954ef8964d6824bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 15:02:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d625e383b47430085a9ed6746268220\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c651994bb26b438ea1bbd5cfc117444d : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/100b81bc14944566aec887345d1f40ab : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ac897bfe7c584f99b6b4a13214eeb977 : SUCCESS in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c214c0a9bb4c0298a63e707ce64d85d568c3f3e2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-10 15:29:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c95da523c81e49e2bcb19537ae9cf1cd : FAILURE in 52m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a076c6c24a164240b5762b7070163741 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f1ff06327b49c7e6f27cccb891a9c08cb2937c42","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-10 15:31:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947663 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":4},{"id":"791c85e9e5f8b32fc0d6231c45bf51ce88cfa1ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-10 15:35:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1400/.","accounts_in_message":[],"_revision_number":4},{"id":"9a463ce421a283bd7450d300b9c87ee2ee5d6690","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-10 15:41:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1535/.","accounts_in_message":[],"_revision_number":4},{"id":"59b6775c6f94776072fc884a31197d47224ad1fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-10 16:01:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 25m 31s","accounts_in_message":[],"_revision_number":4},{"id":"bf21e04011c02253699cdc8c8abb3de6e19a1c55","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-10 16:18:30.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c14affeae4dd05a90e3022f3706f39eced5e0bea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-10 16:22:48.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-4/FC/ : UNSTABLE in 4m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-4/iSCSI/ : FAILURE in 1h 47m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"8952d33620b43a80591df607564fdc90b47dce24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 17:38:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46beccf3940844bda0a5080b1208165b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c652b7cf9767439492df7fd1cf8c7626 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82dbbe90e2c84cba939f3c54080dc5a1 : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6159839fd7e249f5b84a181ac4245ed4 : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e1de2c1e2624fffb2890b4e75efb235 : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0624618c9ac742dfa9fc05dd229d8c16 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/fa85d58d93ba40f288109048496dc708 : SUCCESS in 34m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e5156f256bc48db9f62f813e5807dba : SUCCESS in 34m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5843180e5aa4ccbb868ecde2d0df0e0 : SUCCESS in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dddceea748c4ad0ae5bad40d0317bc7 : SUCCESS in 1h 08m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dab1a8ee7dba49b5a8fa0c24f1e290a8 : SUCCESS in 25m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3367048ba4014bdeabcd31359e5d57b3 : SUCCESS in 6m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f411f9735e5412fbd8c9c9c78880424 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aefb89c5398c4ccd86e17c60fba5b2b6 : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c6075662a337436ca61bb232fc58fd7f : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/185afcbd81944a7cb47d3d9fa35d9959 : FAILURE in 23m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87749b90ff85472e92f8a21e045698f9 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd03057ecb11412d8cb523fc5d976ebe : SUCCESS in 1h 29m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f2ccf13218842f3b29f02a66f143e5c : FAILURE in 2h 53m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1b72b50f5554ba3b34681328d29eb38 : SUCCESS in 1h 44m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bc9b33176916495e8357cd413f60f58c : FAILURE in 23m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/49146642ee794e6083e2f42d51fefe6b : SUCCESS in 36m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/334eb4976fe74aa6bdc2aeb83b6c28d8 : SUCCESS in 1h 09m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9faba14da65f4f9ba53c97812f85cebf : SUCCESS in 1h 20m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e3e1961f7f6d4edd861e9c69e11c4831 : SUCCESS in 33m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4efa587d6a64758bc64ae7b1168bd51 : SUCCESS in 1h 14m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7464e353bba742b4ae38740a24292c79 : FAILURE in 19m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32686d4e1564415aa23cee140b1c846a : SUCCESS in 1h 58m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5917b67eab849c699516ec1e38d2721 : SUCCESS in 1h 37m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db884fa5dc8c4a8cac8d5063f7d1e895 : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":4},{"id":"9ea4c1aae2c7ccd24074046402582674f2a9ab34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-10 17:45:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_08_2025_17_38_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"f457a7428b1ac5915ac6a5cfae0120892390ef72","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-11 15:52:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"26aa7b0c1a70ae84cac6f8f97468f5e877aad16b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-13 09:21:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"852b39949428dd9901972ac4d7153fa3dfe38498","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-13 09:21:42.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/170/","accounts_in_message":[],"_revision_number":5},{"id":"ba1b4bedfdee5ffd4947b704edec76eaf9150112","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-13 09:21:45.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"252a1162c12beae6edff86d63ebcd6a223f61630","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-13 09:26:42.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d5adc3fc95d17b19d65afc28d33c2dc19f0fba00","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-13 09:34:18.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"473f2d4a92c4f9582f77f629f357c89a0c5049d1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-13 10:53:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947663 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":5},{"id":"36cd5b52afdc041fb48514861ca52a9414be7a6a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-13 11:15:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7de2ecbadaec47bb9b0e3c2b469c5cc3 : FAILURE in 54m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/325a8ec3270f4fb4ab6df53109dd478a : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":5},{"id":"70fd99a97eae7eee3ef550451052f8603d568c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 11:19:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b6126115bd242fdb257197e98b8575e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840f2bb96f2c41469ec6922a13f6701e : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365d11810d2c42589401944a68a2311d : POST_FAILURE in 9m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfff27ec3d42403bba101d80241e7bcb : SUCCESS in 7m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39c5aca720d547e4bfaa165b9c0fd090 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8479b03127124eb0ad7e8253e816f6d2 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/0af2537cc8514047a63b6c3ba0d7689c : SUCCESS in 36m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63d4bfcbfdff4410bd13dae3457efd86 : SUCCESS in 35m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa02584e5987491c88b6f4e261011062 : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3698e0606a264e97a6986df151e1170b : SUCCESS in 1h 01m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4a165380b77143688c126307e7c08344 : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/043bc961f7704c81b5722294b3ff6526 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9db92bc170e4c8fa4f1ceb283ac4c6e : FAILURE in 6m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/115b79ea45414f39af4dc45f66388305 : SUCCESS in 5m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f43a553d931b49a6b29e69ad5ef1943f : SUCCESS in 10m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e2d0e0605bf48a49271ee4eda0eb1be : FAILURE in 24m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f7d1a8d04354bb3a9aad78b6d575d85 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/165d842ad7094db3b6a07178734fc0d3 : FAILURE in 1h 05m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/33122699ff2947469ddaadb28efa49a1 : FAILURE in 1h 51m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3541460f5d46482d966cb067d33f1034 : SUCCESS in 1h 56m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74072c8bd6c7456ca77e721413463fad : FAILURE in 12m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/658553ffa27a4bc7b677307f60eef9ae : SUCCESS in 18m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24e6316c270b42ac8fd372df2ac1020d : SUCCESS in 35m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef55c2ce3e5f4ac4a75ac6e0cb55d117 : SUCCESS in 1h 21m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ffe3529b641343deb319b4e0afa45542 : SUCCESS in 41m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/945907991d2d4360964244757b665bbb : SUCCESS in 1h 04m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c7924b4e83104784bbeee243fe218e17 : FAILURE in 24m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7e5f491b469498392e092d012ef308a : SUCCESS in 1h 22m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/887ce092a756429e8a5029b3d2150f1a : SUCCESS in 1h 37m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51d324f3f594495c8290fa587a27b3d1 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":5},{"id":"1f022779d54877162bf89a098614c0f98b457443","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-13 11:37:50.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 24m 59s","accounts_in_message":[],"_revision_number":5},{"id":"bbc62af8769e71b7229f79949c019cef7a6603de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-13 11:46:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1569/.","accounts_in_message":[],"_revision_number":5},{"id":"69c69b14f11a65cbb65872ee4d2b90e5d67c9570","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-13 11:49:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1434/.","accounts_in_message":[],"_revision_number":5},{"id":"938999915b6905d432ee777dfe28d4f002c764d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 11:58:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/889d93cce3624e61bab5b90aa9dd6352\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7d5f02a96b7e4de893e7c9f79c388002 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0c1c285ba6c4bddacc8202dbaedf089 : SUCCESS in 29m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7467418d2e904287826ef92e32f90b63 : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"508dbebe2d4f5029a845718702263b4f0947d372","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-13 13:01:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"324d09b92e736619dadc57aaf3e1e0174076ae85","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-13 14:46:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f632e8c2eb454a6088140fb8c35ffece : FAILURE in 18m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1e15d549012c486e9e8cae941f7a0070 : FAILURE in 1h 00m 14s","accounts_in_message":[],"_revision_number":5},{"id":"77bc707fda3a8aa17b56c5a9365f41f2a921c71a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 15:57:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4a4f925136476c92c73297a82ed70d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d180a37f959c451c8c07104f8e8d48b9 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50e0dc619cea48de8d160b1a5810859e : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe6ca3c816084eb0911f2ead7384439a : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4d665ac5ed5a73ff70d27d72c61289014687d532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 22:04:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0389d7b2c8a4c58bcfd1687016b23aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e5b916c9aa401490ce96128a660453 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3e80d275f4944719a9172ca932b7380 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b76626288d4543d6add397cd30148e92 : SUCCESS in 12m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04278bc9dfd641a9ab42554c1fadae01 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3db996a4faf472aaa78b3a4835c7eb9 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ceb69f52ae8145fa8b2234a32c7cec99 : SUCCESS in 34m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18ee2a2787cc45f1b91a6782a1da113b : SUCCESS in 1h 01m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ae788c9c7ec4c828fbf889f0ad2e7b7 : SUCCESS in 1h 36m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe24f94294b54e3ea7dc7b9ce8536004 : SUCCESS in 32m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a8240eeb38e4771815b78b2eff04f02 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7b3b3e713e474550899396e2d4be31a9 : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c48007a6e76545d5b3210861e3568acc : FAILURE in 7m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f489b0208e904fba938b9d2e34a738d2 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/532a3485dac840b6970f23cbd3cea41f : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/75cb25dcab9741f39db7c4352a1e5210 : FAILURE in 31m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/687f9fe1c4ca494cbb1e6f33f4ec1156 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64bcb095e2aa417a8282a97e9ba26297 : SUCCESS in 2h 04m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce26d508b0124b758e449f03ca6dc1dd : FAILURE in 1h 41m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ead6ee06e0d4ae9b885af4e003df7b1 : SUCCESS in 1h 43m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fc5b2bd56d414baabc7fe9e5a1d25e1a : FAILURE in 18m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c07e87eac1b242a7b152b7d709e452f0 : SUCCESS in 37m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4bb28c6b7844499bdc28e836be86623 : SUCCESS in 1h 08m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f8efe15f3844860a2d0057277189416 : SUCCESS in 2h 06m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d7d03cc12934984b94ea9788cb03026 : SUCCESS in 43m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8682f9f7c692415cb8ec2644333dcdd2 : SUCCESS in 1h 41m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9fb249a31add494eb4aedfb323ade831 : FAILURE in 22m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/200fcb1bef574e1780c1013c991bf39b : SUCCESS in 1h 34m 14s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeb44ab9aa164efca5ae45539072ff0c : SUCCESS in 1h 28m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/017582a7a0024a6490648a59d30c4612 : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":5},{"id":"62129a66d4bc86e3cefdac66e727ddd747c41740","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-13 22:10:09.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2025_22_04_45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"d231fe19acea4260f532c5ad07948c688888cb51","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-14 14:06:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d7903bbbfbba0d5ba5bf8a3aba3fc47b7cf5ee0b","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-14 17:39:25.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b87c12a1cda2549cf192533631367caa5cd7dde2","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-18 06:03:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6c4ec0d62f856b7ea5a837352b743e45e9729777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-18 18:17:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"786284d7da8b03866058126e76d62911ea9ceb2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 18:17:49.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/201/","accounts_in_message":[],"_revision_number":6},{"id":"998481b9666c77649fb8d2f10b2a016c6e6fd111","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-18 18:18:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"fb21ad3947255895a49096ef1eadadc99593e487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-18 18:18:50.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"2b4c86bd69f2c8ba980ecea540e9068683b719f7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-18 18:19:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"b4fa2ed59f2762ad44b0b4dd2b2ba1ad31a0e85a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-18 18:22:00.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a105f98b7d8c682fd95cf78759c4c26779d871c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 18:31:59.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"00f45b8ad64d914ea431951f772e10531ee047d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 18:32:00.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/202/","accounts_in_message":[],"_revision_number":7},{"id":"a1b1c0e9b7dda8294dc7b3b06f5e751c52abe9a4","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-18 18:42:29.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bbb412a1cd14cf6a1c017e1e92739c6aa32bffdd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 18:44:28.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"e8e94877e20184004a080bae078455186078b2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:10:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bcc43e804e44638aff041c62e793477\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27e73fd2f2d744db82e41aab2d7b3053 : SUCCESS in 22m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b862fdc2d0c420e8f4e6f7b2834a915 : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1393a88098f34d639131d4108ac0ac6a : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1545b46b0c13d64d94b527afd129bb61d44e4314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:11:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de209be6377444abb8d093ac0c942c26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7bc27a8a6f74bbd82de4aed7669bf5b : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa2d53d85fe2435d9e4b560a4724ed79 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3be07c92e76a4de59ab420fe6cf9b8e7 : SUCCESS in 12m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff6de43ee8e54243b75f118540b8a835 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8771f518e04a08b62ef53fe09973cb : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/dab7c00cc8ff44a29844e0953e25f734 : SUCCESS in 51m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83eab2945b9f49e182faec9e1e854732 : SUCCESS in 51m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94b67535d5f849d5ac013560e45d410f : SUCCESS in 1h 33m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46a9d7c1e7ba447b855d28ea4343d7ec : SUCCESS in 1h 03m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2959c49cc48441efb639c83005ce7bbe : SUCCESS in 13m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9e036803c9643c288698d5b588eae55 : SUCCESS in 25m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ad5807b12bb4eaca405c16d7489c394 : SUCCESS in 6m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11c960e6f57e4a258761d1b6a3d09bf2 : FAILURE in 9m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/672492b8c7de46f4a44a4083ba3193fb : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f3b5932337dc46f58e3f6cba5d4cffd1 : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/156d7fedda3d4fb58a7e28baf87b6978 : FAILURE in 19m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc5c1a3294b84a08bfa9244a685e06fd : SUCCESS in 7m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac6110c6c36b4cec80796eec4bf67ada : SUCCESS in 1h 45m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9babb274694e4e248ea63fb371e3edfb : FAILURE in 2h 37m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c35c487872be41edb7c58e63da0c6cb5 : SUCCESS in 1h 56m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3a8311b63df04bfd885c4bf1c115bc01 : FAILURE in 11m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/60b3ec9c7e8447c5bbf262d24289f124 : SUCCESS in 28m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db6cf52b07f247ad9b542b4b744881e2 : SUCCESS in 1h 04m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67212bf1d1064322bfca87c092038dcc : SUCCESS in 1h 29m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eea63548509e4bd9928cf646567c3558 : SUCCESS in 40m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7f45f61cc33f4ff3a7a46abad794ee99 : SUCCESS in 1h 21m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8f19194aafcc4e64a5a0b2135e3027b7 : FAILURE in 12m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3921bf704e4e4197b1fe36ef37b0e660 : SUCCESS in 1h 28m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/975ba9256a4645d8b6252e6222141b88 : SUCCESS in 1h 28m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6037b27455f141319888490921261d0e : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":7},{"id":"fe1607e7f61d13d81b701857ce61362a66410e50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-18 22:30:54.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947663 : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":7},{"id":"d43aa21c51264fbb71b8878e115f70004f2b4a30","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-18 22:45:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/40be775575ed4383977963d2752ccae5 : FAILURE in 47m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/042ac3f49a1b41b39701c434c3aaeee3 : FAILURE in 1h 06m 17s","accounts_in_message":[],"_revision_number":7},{"id":"7c88c73dd0b14a8c593ff075dfb11fdc9656fe16","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-18 23:04:09.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-6/FC/ : FAILURE in 4h 44m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-6/iSCSI/ : FAILURE in 3h 47m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"3e811b1cb2861393d725b827d54108f8031e4714","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 23:08:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1604/.","accounts_in_message":[],"_revision_number":7},{"id":"419b716f621ec2c8cfac9862aab7eaf17fec8485","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 23:11:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1469/.","accounts_in_message":[],"_revision_number":7},{"id":"4f337e8d5e9e76ffbbc674455b06895f40da0ac1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-19 00:14:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : TIMED_OUT in 3h 00m 37s","accounts_in_message":[],"_revision_number":7},{"id":"26d5fd75ea634e9fca2b1b0e64ec536c11c420cf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-19 00:40:06.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-7/FC/ : FAILURE in 4h 54m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-7/iSCSI/ : FAILURE in 6h 20m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"eefc5bf16279344caf95de93a6b71352bb8ef081","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-19 04:05:48.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07c228c9b951552c8394670330f3cbed14844b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-19 13:32:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"60e8fec2e549eefb740e709f9c97c2eeb789cf2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 13:32:17.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/216/","accounts_in_message":[],"_revision_number":8},{"id":"577765e9ee35da6ca448260a79b3ba32100aacb7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-19 13:32:19.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"f5a730ccba4972f41d9673d5a251ba910dbba005","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-19 13:33:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"992fe0a130797a203b2720d4cb9c785229ca6f0a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-19 13:33:41.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ae4b3eda57a12bd15d6519f61841dbde06112027","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 13:45:56.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"eb763bcaa3365851fa95c8b89b9375338c5beaa9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 15:08:47.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1617/.","accounts_in_message":[],"_revision_number":8},{"id":"f84cf9426a608fc52e3649e4f39ddc9462cdb042","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 15:12:28.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1482/.","accounts_in_message":[],"_revision_number":8},{"id":"054e0d1eab2c6feafcae9560715debd648a4618a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 15:58:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c475db7dab134c2ca2b921b054c0f628 : SUCCESS in 48m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bcdab17756de4f188de608e6671919d0 : FAILURE in 50m 27s","accounts_in_message":[],"_revision_number":8},{"id":"df5749ccb13790e7d970ba9b30e4fc5669de8048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 16:27:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34147bf90bb44fbf8fe2cae7b077e1d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c074ceb65a614e47b51cc393896806a4 : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f35672ddfd34489db30ce6151438482a : SUCCESS in 7m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c210c532445c4d58a411ae6eeeca62bb : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da352b96e9b74f4fb9e80e449ade1551 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485831cd9ec440cc9b8f7f105aa5789f : SUCCESS in 6m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e3c943eea2ea4b328bc464d3973f3f18 : SUCCESS in 46m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a77ce5d05744ebb8af06ca7b2b01fc7 : SUCCESS in 57m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a274510f42644e6e92a971c4d03d60f5 : SUCCESS in 1h 18m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d0d9c07db9e4b599a6b5efefbb4c884 : SUCCESS in 33m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcca5fc43fcc413eaeb604f8e9ad5c4a : SUCCESS in 15m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5551c51aa9884e5fb5a7f0dd73376b96 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dcc2b938f71645eab6abb17ab449c957 : SUCCESS in 3m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b6998aa88f748378c95b78cef7412c9 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66efe68eb25445f982301b6eb493eeb3 : SUCCESS in 4m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2dce28005d224e3e9976cbdd92e36d7b : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce7da7dd80564ffbb823de5287711410 : FAILURE in 31m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d88d0f2eb354aa1bc95ed7b71b0c2fc : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0f01e7ef14f742c1a4c42b363a40dfb9 : FAILURE in 1h 20m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d396135f7d9d442d96c9f5b93c9d86a5 : FAILURE in 2h 38m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f326f46bc6749de991556ec6ffce782 : SUCCESS in 1h 51m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/94165a4b228b40ab826887b5595f5a55 : FAILURE in 14m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7564665c49264e5ca8be68388313c9ba : SUCCESS in 16m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a9ffbe0623bc4b47b848f70bdc81f16c : SUCCESS in 1h 08m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f8fa095b33f4a12a280fb5d1064f81e : SUCCESS in 1h 41m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/004944aac0c54827a1c88aba8e73c614 : SUCCESS in 32m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ef8c80cab9442f7957bf3ef74686129 : SUCCESS in 1h 07m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/98e0633ad393456f8de96c00845069a2 : FAILURE in 9m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea9b998dac8349e28d6766a292cb2110 : SUCCESS in 56m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8bfb072f1a14b85a4294d29ef33b6ec : SUCCESS in 1h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97e59c229dc04db9b6aea452504c8a9f : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":8},{"id":"3d65bca5cf15e0e92cbb9b49a0e5cb50a33dea9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 16:46:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d5d799af764677b290d6c915ad2cd5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b927ca6cf1c549c3a4a5f5720dbc91f0 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed93a16b2646410aba7b4c2f6cd67b8b : SUCCESS in 28m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f0e640223764903a8d1a14831ad3660 : SUCCESS in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"18aff708470d37867133dabce30137c1148dd94d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-19 16:49:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 33m 42s","accounts_in_message":[],"_revision_number":8},{"id":"9d7268bb8739e577ff9053f696e163fa7d5d6043","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-19 18:58:34.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-8/FC/ : FAILURE in 5h 24m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-8/iSCSI/ : FAILURE in 3h 44m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"1227e577d499758c381a62c5e39842b6fe4c979a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-20 05:04:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7dff1020faf66aabb5155febbfee565065ae1f82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 05:42:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b4671f9e77944a0b0b93a42f6bb0842\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1f8944d3fb34ff1bb91c967f00ca52b : SUCCESS in 25m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bccfd523f0ee4feb8a3c4b340e71322c : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13fb557fe9b041bfae8f8bc26178c920 : SUCCESS in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dd8813b8f34f62ca69f8782b255b0eb8a2243f57","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-20 06:00:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/23136a6a02804ced8f13ca3c1cff5aad : SUCCESS in 54m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4da02c07b06c42dfbf3482887a2187d1 : FAILURE in 53m 39s","accounts_in_message":[],"_revision_number":8},{"id":"7d8bc7ba18ec438b08aa8d0b1232bc1f7b0e2b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 07:14:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa5b52a740d04fcb82c3e70c5e8ade13\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6d949af34d46ab9601533e2912404c : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eab333aa078e4416802b8ccaa11c49fd : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c37f29fa379540d6be81b99d9a23e14c : SUCCESS in 8m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee7b7db9bfee4812a0b08b0edf1d3b80 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c121d8a0df3a400ba03668451d6d2c86 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3c85c1f76a4048d1912d6dcff1b63965 : SUCCESS in 1h 14m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c52a1549f632463aa8bd26ccd69839a7 : SUCCESS in 51m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c77ec88e438b48c58d1c63528e7ae44a : SUCCESS in 1h 46m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f21f4c3bc00245bbb545b0d080b90d85 : SUCCESS in 1h 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/944a8b3cf7e3401187d45ace380c4cab : SUCCESS in 16m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60c3be183b1e43fea370090b47634ecb : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7786063ae3fa4b1ca0513d1496740276 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7bd6767141eb450490206e3208cb12c7 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11909fe238a748f3be015a2c05b6eed9 : SUCCESS in 3m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1a8096cf0f904e6990f25e60f110c3ce : SUCCESS in 10m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89a6b60994124d428c1204abc00240f7 : FAILURE in 37m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8ca8e4f5a8340dc8a124e52f33d1bcb : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8209d7eca58e49d1a93e5e094ba7ad7e : SUCCESS in 1h 51m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d92ca9e47ef4d7bad782d27274a39a4 : FAILURE in 1h 23m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf1b4b2ffd8f4a9ab400a50b7a71d9f4 : SUCCESS in 2h 07m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4ff38f1826847659eeda2a9427a4caa : FAILURE in 10m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4210810c1ef94b6dbde48fd41506502c : SUCCESS in 38m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f54bf2164db48eba01d0738457e8952 : SUCCESS in 1h 34m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cf2a55f97954d90918a6a74ab673f3d : SUCCESS in 1h 39m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2043729c9ff547b0bbe215b3f1fae5a4 : SUCCESS in 45m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a76f835544844a8fbb5bf97e1e7bcc59 : SUCCESS in 1h 37m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a380e2b25e3b4193ae42114a9dd94867 : FAILURE in 18m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5cef882744444d89a1f4b3f7387b97ff : SUCCESS in 1h 27m 04s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b0556308c6143ae81f43418bb34421a : SUCCESS in 1h 36m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad8236b1fe4846148a5e22ed37ab114a : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":8},{"id":"1f01fe2ed9167e89062aacde763422138b8ab9b2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-20 15:02:54.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2c4f87ccd96967da655315193d85bbf12c172c8d","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-20 16:19:56.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f00b89c710b272a7e6f6772a51418a314751833f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-25 07:54:23.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"65ac41c73838ba22d963b3d5798517a3cbe76f87","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-25 07:55:09.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"0635b14c1670e1ed364d163947090402d38f1f7f","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-25 07:55:53.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d5fbeee7078648d9d513e12d26ffc6508a267d30","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-25 07:57:19.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1e87362c89ef48d0a1049396c2a27c65 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f02f9bed33c94234870c2621715f5880 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"b67a90541616508b62149c94a239a10f6661fb4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 08:20:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a5b18e2c094177a3ee26cea16a583e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3c9b034c7db4bae877a9fc707d4f0bc : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90a62d474d9e4080af1acf58f7a62893 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f79ae936c82e4493b857dcea814beeca : SUCCESS in 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5782122f14c77d840937a4a82bc74ae91aa38de0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-25 09:17:37.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/282/","accounts_in_message":[],"_revision_number":9},{"id":"3dd92968a91f7c5c2fad931b6507267e6f098159","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-25 10:11:44.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"c3779376a6234676af7e13b6528263b5554859b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 10:25:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9810073b36f4621ba3e12102886cfa5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5053f3a33e457ba6bc96e478c5c31b : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6929d25c4034154a4dd378a0df4ee69 : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bc3cb8ef7a54dd8a2de11b44742a2a1 : SUCCESS in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dea1921b044043f3b4e5fe9edd178aca : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e8f6b3d8254c6591e548183f87d427 : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/941d65b5360c4086b74289b5a4e49178 : SUCCESS in 31m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da4c1001ed4041fc9df4570befcfc801 : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25af0e3947414ef982bbd8f284e7d35c : SUCCESS in 1h 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d1c8b74c0be4274a2e183a1c5f6d479 : SUCCESS in 1h 00m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a65b486be1b48548377ce15acfc67ed : SUCCESS in 20m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f9735398314443b2bcff4ac6a0ba8c53 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f571dac5851d4ae7a222fd5b3a56b72e : SUCCESS in 3m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba278cc0793e4816a567a1fb1622ebc5 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d504d91515904560ac4cbd710d77ae45 : SUCCESS in 4m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/822922ee8d134c30a6ca925fefa89b39 : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99ceabda087f4c14a17b5f07186af1c0 : FAILURE in 20m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0c94eb877f74cd4ad2e253baa751b62 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67a957bc07554b8992d40f2ec90a5b0a : SUCCESS in 1h 43m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/572c3e15c12e4a409d861876496072eb : FAILURE in 2h 25m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7671d7f7235844dc881a846382a5fea0 : SUCCESS in 1h 26m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/98811217130f4e8bb7b0354618c90f9d : FAILURE in 15m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/36b427ebcd9b4c1a92f2a92c4f549d74 : SUCCESS in 38m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9901b8f7f5c4dcabb1fc549428df30e : SUCCESS in 1h 07m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85654cca474249e49118578ff4f94612 : SUCCESS in 1h 09m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4697075e97db499896391a416e8067c0 : SUCCESS in 43m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6f13b2bae99542c499c2690a9df13493 : SUCCESS in 1h 41m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2a4ba3d2042846078e4f1a153c083b3d : FAILURE in 18m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8dde99ca69784e9488084e080a6959b3 : SUCCESS in 1h 29m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f1c1616ee29431c9f650c19b6710567 : SUCCESS in 1h 26m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/471dc00c76524fde93b195e8eb12868c : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":9},{"id":"c1b68f857f1a62d635b3ade27a99d7243b3ee421","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-25 11:31:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 26m 10s","accounts_in_message":[],"_revision_number":9},{"id":"59b813b87c61fb4e60ba86606d9a0fbfb6e8efa6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-25 11:47:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1548/.","accounts_in_message":[],"_revision_number":9},{"id":"f66d7800c0cc5332a5dacd264f5987b35e9325b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-25 12:02:37.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1682/.","accounts_in_message":[],"_revision_number":9},{"id":"fa6c29871361ae2f1c971c82015682bc9046e083","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-25 12:44:36.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"cc5b7f378be0fce4f72ac99540cb3c7e2ae00f8d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-25 13:09:05.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-9/FC/ : FAILURE in 5h 13m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-9/iSCSI/ : FAILURE in 4h 59m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"3b34dcc9f7af9cd12318a4381817f167724e7ca5","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-26 18:08:58.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cbc3cc1a1564d1030dbd3a18fea3f8261a0ab38f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 20:55:13.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d0eb97fd39cc0bae489365df0c7e1e9d4a37d7d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-27 12:22:03.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e83146fc2ecbbe6a042037bb402f77f40427d188","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-27 12:25:47.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"3acad768d03b49c1a61260f4c14b9dd854b3f4f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-27 12:34:50.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d4722f28423a4ecbc299df668b1130615a05ffd1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 13:30:04.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/323/","accounts_in_message":[],"_revision_number":10},{"id":"16cf3701eed9f4a94672c33151691309ff164983","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 13:38:21.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"4ee44cb9943b16b8c4cec64b88a91e6c5c9a3c5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 14:23:44.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"71103962d6035cda86e7230da05a309422740975","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 14:43:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1585/.","accounts_in_message":[],"_revision_number":10},{"id":"b02a08b11241dbbbfe68c383f41074b4276a39a5","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-27 14:46:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ad8056dfc5f49f95cf31e6e05af76d590e6879e3","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-27 14:47:13.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d832dad5c10bc5b2473adb157ed000ddcc896857","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 14:48:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6ede48a9a5ad4919b3640296fcaa2eff : SUCCESS in 44m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/10ef5b17da6446a4a1ed5ffd92c75e2b : FAILURE in 1h 00m 12s","accounts_in_message":[],"_revision_number":10},{"id":"3491c8d5e07dc3f11970f384744d85d72dd14536","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 14:50:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1720/.","accounts_in_message":[],"_revision_number":10},{"id":"9ad6723821f69cfa4fb0dfaf0db1d1bb81c6c117","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-27 15:17:43.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 24m 43s","accounts_in_message":[],"_revision_number":10},{"id":"4af78ea15d0515a1a997daa133df7cbb975a7e0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 15:19:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11ff026e0a9b41b7a43679cb0e356d4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675999b4aa504a77accbe25b48f0ebd2 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20f6dbfe98b34ca58ccec14e7f05431e : SUCCESS in 8m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a15730e248f348328291a1640b3954e5 : SUCCESS in 11m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a529c8844f1a45f388f2ecc136b1de6d : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7853c7e353fe4f21bad7001b52345136 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/92d2e886b808435fa7c428cae761d9c7 : SUCCESS in 27m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7587a03b7e5c40ce9c0ddc90da57298f : SUCCESS in 56m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5d167e453ed4b1f920f103f5836ab63 : SUCCESS in 1h 32m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a67a2f427ab44bc2b82d74500a6b9988 : SUCCESS in 1h 01m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a4b0f06bd644fe8dc1b5acd14b58ac : SUCCESS in 10m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9231a80396f249d78a5b7548da7a8893 : SUCCESS in 16m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/acc0e188458c4c97876e10ed433dd61e : SUCCESS in 3m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31ec3d796b9646c4a9b50cc1754a60ab : FAILURE in 7m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f156c28b66845bfb9a602cef6edcbf8 : SUCCESS in 6m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/44c3a7b5e6384a30b5249fbcbf61306b : SUCCESS in 9m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/20e9c8309fd9421f8561d3c5322ffc9e : FAILURE in 35m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4897d34c91c445f0adaf7ffd75861901 : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2af46165e702419fa577f94c9407cdf1 : SUCCESS in 1h 45m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9c15ca90c6c9453881ff6dfcbc1e7095 : FAILURE in 2h 51m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b2e70e795b864e88a1d609f618ef2d16 : SUCCESS in 2h 02m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/76f83fb91a1c49eaa1fb49875ab12618 : FAILURE in 18m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/93b3250087d84d0488d831ecdb496f47 : SUCCESS in 38m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7a5f37b812744548551f156779b5252 : SUCCESS in 34m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e6cab6bee7f74a8da17554d35f756389 : SUCCESS in 1h 43m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19aa0708c229434caf1b1eff374e91f4 : SUCCESS in 42m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/97ee47b5db0547648523fcddcdfd4e01 : SUCCESS in 1h 33m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7c6d66f61e374525962bf59e5451e764 : FAILURE in 20m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f674d1ab2ef4adaa0271e8d8fca8f97 : SUCCESS in 56m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a4ba42f2751449292b0a43e8cf7de95 : SUCCESS in 1h 24m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/660b242013d246fa91b0e7e7c73973ac : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":10},{"id":"9a3e1b5e138f6294de1100e8f32fdc80035fb200","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 15:57:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd2c1aca03349859d8fc3713c62fe8d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20c474aff72548d09113b721213256e7 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/590133dec7534850ab201baa49ce86dc : FAILURE in 30m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/623444eac13c497780bf98939f266723 : SUCCESS in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"347bcf7e56497097af9df76a4c4f37918b9391be","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-27 16:00:33.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"37f15ee6d2100b4a3f380251d3a58b9ddf5135a2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 17:25:18.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-10/FC/ : UNSTABLE in 2h 7m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-10/iSCSI/ : FAILURE in 5h 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"d5d057c3d5653a40b7a18047ace5d977f73e6204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-28 19:10:38.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"2e416ff77b5223620e79fa73cd783e952b133f80","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 19:12:02.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"7408ea1af185b589097fa59a4f158eeb26a9c4ca","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-28 19:15:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"929dcb0d9d6f927805a41950850274c26d25167b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 19:17:18.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"731edbda340f1aa2a50b18682993d145f7a4ec83","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-28 19:18:19.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"dabeb936858af80f1ed08b7d4efd078a9f807229","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 19:47:39.000000000","message":"Patch Set 11:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/340/","accounts_in_message":[],"_revision_number":11},{"id":"00c032d8128e34ab165329ddec5854465f88585f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 20:41:43.000000000","message":"Patch Set 11:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/947663/11/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"aef8072bce436897b057fe731846ebac7a53f481","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 20:54:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1600/.","accounts_in_message":[],"_revision_number":11},{"id":"b155d81fd6b56971d024e756140644e4b62e92da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 20:57:45.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1735/.","accounts_in_message":[],"_revision_number":11},{"id":"0669e2b56799a195128fd962eb0717d6c7d81394","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:21:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f174ba251b8e4b5f87aec1d415a2ea59\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10603585f9404a7cb185410e9ec120b2 : SUCCESS in 23m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81ef5995d9fd4ee3ab2782553a2f8d14 : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2301ccee3c34be5abbb058d0c6e4667 : SUCCESS in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a8139b24abb9a5f43fd13dbf5e876a77c6b94d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:39:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec8e8b89cb764503b5e660d712e51861\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f518e8efacb14b12a3b77ce0fb22181b : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2a659513fb74305965d24ea945eea1f : SUCCESS in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b1d51715cf64765802328171e5aa79f : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b99e2aa10af2406aa826da563adba712 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51b0f724ca5415685ef9b045edebf9e : SUCCESS in 14m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/347ca79349364cebae13c2c92d0ef32b : FAILURE in 11m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14ff3b6072944b0ebe7eb76f15a2a9f8 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb61b3024b2f4abca81814be442fbe0c : SUCCESS in 1h 34m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e23194c1cba42278e1cae231c685ffc : SUCCESS in 1h 06m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55692b08887348a5912ea1a8dcd62721 : SUCCESS in 8m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c8b289a5e374059b19dd684e6e232ee : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/35fcb40ea27c4ce382128b8e9a81e3a0 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f1069f662e141c4a0dd5add84454183 : FAILURE in 7m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/702afb1ba2fb4e03ae412c3068c5637c : SUCCESS in 4m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e434723c3fb4d71b94c8d045b3779cc : SUCCESS in 8m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c0385a3b63040bd8d762d6a6077e83e : FAILURE in 32m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f6bd845af1f4295aac8ab90a37501a4 : SUCCESS in 8m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f07904672f38467f81463779738eb32f : SUCCESS in 1h 09m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e1b9669a7744fae8a5821af7ab077d9 : FAILURE in 2h 08m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30c63d0d8efd4f44a140c6c5585a1f3e : SUCCESS in 1h 28m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a1e08d2916b541b084899364eded7c90 : FAILURE in 12m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1859d7d19f314a7b9118073afbd2b213 : SUCCESS in 31m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/75227619dc0c4c36ac589b8229b1de85 : SUCCESS in 59m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/414c09a008594be28a0f54a12eafe323 : SUCCESS in 1h 33m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/03c5369673f24cfbbce42775dfed5360 : SUCCESS in 32m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a1fa90c5b614bb9acf16644207f942a : SUCCESS in 1h 07m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/043526e99ed944edb3a5229c0a13992e : FAILURE in 26m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c0d02441a8440aa933270c4249995d4 : SUCCESS in 1h 18m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2723c16eed841eab243064e61b8b0ec : SUCCESS in 1h 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9024bf889ad84fdf870372c6dfe95552 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":11},{"id":"107a1956ded84a7297465f1f2e2556bdbe6ff852","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 21:50:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947663 : FAILURE in 1h 24m 30s","accounts_in_message":[],"_revision_number":11},{"id":"6e889b43c6e05c5da14a3c093154e280554766c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-28 23:32:32.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5cdf47c45476441a8e8445c20a589435 : FAILURE in 1h 37m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f34ec6a88d514320a423913c1ca7d0b6 : FAILURE in 1h 47m 16s","accounts_in_message":[],"_revision_number":11},{"id":"5c4c914875a3b31e500ecc371385d49eeffa85e5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-29 00:40:15.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-11/FC/ : FAILURE in 5h 26m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-947663-11/iSCSI/ : FAILURE in 4h 2m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"bb9a73cbe2472da79aa7c076df8a6606ee1f0d83","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-29 04:22:35.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f4126375e092f942c700ad7a1a9ff797b513df14","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-08-29 04:38:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"604ee512285ea5595eea92c2ab2477d842551a2b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-29 04:40:12.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"3d0dfbc4baeea6d75ac569fde8cddc97199d6c75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 05:05:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3620e6b70dd841ad9beadc63896a5302\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a93e059477e8407fad45ac5decb1fb93 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/225148805f3a4454aa2152deaefbe37f : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89ce6f2da36a434eab81efb11759893c : SUCCESS in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"957a3a6c221ee6f44e8616b6d96048ab17870c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 07:48:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef04931966f34ed6941e3bd316fd3868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d0ef6feb3d4baa97e7ca96bbb86c3a : SUCCESS in 21m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95013ee11dd345cdb5f244e7d6e68d3a : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/675f422e010745ccadd9e680b24057ab : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20ab38536efc4b68a6eb75b40d0d9b7e : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746bef7d80004a9aa99d0b1ce6309ce9 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/7b4f3a6e004f44b0a0b4f15c7db65fe0 : SUCCESS in 31m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9026347f800647628536109e210f68bc : SUCCESS in 32m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/070b8cad937447c4909fee0a73b84023 : SUCCESS in 1h 17m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17d3a830f2d44fbc90ca65d7b6e6ffbc : SUCCESS in 1h 02m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89bf05aa6ce344caa829141bb2ca1179 : SUCCESS in 17m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd2a3f8e0a8d480ebe3549f0d73c5377 : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/53c4bd4722da4e7799a92d38d7949b7c : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bdb266c6f4eb438587f86c8fe07a4726 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bc011fa34ae411cba932b2cb6f43ae2 : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/43b239c245a3425a96be83f79fc8f573 : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c8e4df6fc5343fa8c555932a775de14 : FAILURE in 50m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c0dcff89b9549c89d6adcf89d14b32c : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b29dd7fe53fe4733844f023559c641d0 : SUCCESS in 1h 36m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48258092652846ccabbd71b47f7cd7bc : FAILURE in 3h 02m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d5441d31ce44ca2ab608aea1b1f1ede : SUCCESS in 2h 08m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8bdd1d4806954c1d92a6e4a3cedbdef9 : FAILURE in 18m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ab57abf374da4acea443df4af9f3cd8f : SUCCESS in 36m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9ac1fd18cc8f4910b5a8708e95e7c829 : SUCCESS in 40m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1cbfbaa256f04b65b6eafbaee9a215da : SUCCESS in 1h 42m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d55f7143f3d47749b17e4c2618c50f8 : SUCCESS in 49m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/83cd4a2788d4472e92b59e2f58ffc6db : SUCCESS in 1h 38m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/32608fe734b940cbb11fd5e8886e6e80 : FAILURE in 19m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/087d2c1188fd43b3a6568aea9d216d15 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/775f29f9ca7142e9b76e1869b741d8ff : SUCCESS in 1h 31m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ade4e9ad90b4580a42518a0a72d0412 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":11},{"id":"67ad00088169fe609e41cd0aa6e383085f74ff0b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-29 11:51:10.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6258dba7bd9ddff23b1479bb434fa21bde73134f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-29 15:08:54.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"19f329a5f22c4db9666b0c1691faca429185ff04","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-29 15:09:59.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"50a5f264a75db80825e15c767a349714db98213f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-29 15:22:19.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/27345a961dd44e5d878a5d7c9573573e : SUCCESS in 1h 44m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aa2deee401fc4ec8999407cff8d2011f : FAILURE in 1h 32m 55s","accounts_in_message":[],"_revision_number":11},{"id":"14c7dac9438e3af7b4ecc5ba167885ca958d615a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-29 17:54:44.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/947663/11/upstream-check/cinder-tempest-plugin-ontap-fcp/7e2069e/ : FAILURE in 2h 44m 11s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/947663/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/24c2137/ : SUCCESS in 57m 38s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/947663/11/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/35c491b/ : SUCCESS in 47m 05s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"1c754c6bd9cc38d0cc45beae30406b41d3402ca4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-29 18:06:02.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bb39ecec8e5b90f0923612cf926e6c087b9183d8","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-29 18:06:19.000000000","message":"Patch Set 11:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d167f24de7cfc6793d2585c702701bba3b61acd0","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-29 18:06:52.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/2c17406dbbae4b94b9b3676226815d37\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/abe965baae3d420bb7a361b9c2b1b809 : FAILURE in 29s","accounts_in_message":[],"_revision_number":11},{"id":"393abfab515d07b0607c7a44c347b7c9f99ffde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 18:06:53.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"8f1fa26bff9e0686d28fb35b2781234c8c045980","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-29 18:07:00.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e8ee20b8c92073f7929b4ee371fce62751938166","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-29 18:07:15.000000000","message":"Patch Set 11:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"22bc05d21af61aee3b7265415089d3aba2b9fbb1","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-29 18:07:47.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/cbf257cadd8f4a60b5d404ae8fcf90a5\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/2c28c40242b94c3f918834c6d2ffd151 : FAILURE in 29s","accounts_in_message":[],"_revision_number":11},{"id":"6ea119c7e30239d0099305c70e972c7d2906ebda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 19:49:47.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d1a9cdd748d4e00a39cd08e99a9829e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08813a0ec2d141b2817df5bc360ee688 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0897ae1ebbbb4604b26e8ed71cf61fd2 : SUCCESS in 8m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb7b9071b1444451a33058d969d93ef8 : SUCCESS in 16m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1583ef7bf7041968ba0597d31764ad5 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a95ffd88897f4d1db2ec95e1726c0d12 : SUCCESS in 29m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/875ed3b067194ba4b3d32a644a2245b7 : SUCCESS in 49m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75d0361724bd4530be9b8ab4151c4b96 : SUCCESS in 1h 07m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4ca2dc7292f40dbb8947f5e174c98a2 : SUCCESS in 1h 03m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54ac8960b96a4e2f82abf81cd534bd9a : SUCCESS in 16m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d493a9907a14a50abbda762fe232a1d : SUCCESS in 1h 17m 07s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/692bddd911874eb9b8847dc2888c4c0d : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e10bd035791743c4b9d6749bb3a16d11 : SUCCESS in 1h 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bf860089363467aa7c58dacaf81f4c1 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":11},{"id":"10409bdff0447243ed22332d34f88a468a4b8620","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 19:49:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"e6ae39673feebad8ad9bf30337b3897ff9fefc00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 20:04:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997a4f78146349f0b4c72a98049c2663\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebdf466851d412099ef2d77a9cf7ddb : SUCCESS in 5m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ff96cbcf96c40f0aaae82f64f145e88 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d3d91d9a13f8266c3ac2489859ce14545792bd26","revisions":{"0833f70efb12c73596de8f0088a7cf057b6c91a6":{"kind":"REWORK","_number":1,"created":"2025-04-18 12:24:41.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/1"}}},"commit":{"parents":[{"commit":"d52e0532f978507a742504d4ab1dec3fa8473b4e","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52e0532f978507a742504d4ab1dec3fa8473b4e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport Current NetApp Cinder driver is not supporting https transport using self-singed certificate. This patch will enable transport using self-signed certificate.","message":"[NetApp]:Cinder support for self-signed transport\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0833f70efb12c73596de8f0088a7cf057b6c91a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0833f70efb12c73596de8f0088a7cf057b6c91a6"}]},"branch":"refs/heads/master"},"d3d95ec459f6faa6fc21345d0be2aa02974843b4":{"kind":"REWORK","_number":2,"created":"2025-04-21 15:22:45.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/2"}}},"commit":{"parents":[{"commit":"d52e0532f978507a742504d4ab1dec3fa8473b4e","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52e0532f978507a742504d4ab1dec3fa8473b4e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-21 15:22:16.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport Current NetApp Cinder driver is not supporting https transport using self-singed certificate. This patch will enable transport using self-signed certificate.","message":"[NetApp]:Cinder support for self-signed transport\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3d95ec459f6faa6fc21345d0be2aa02974843b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3d95ec459f6faa6fc21345d0be2aa02974843b4"}]},"branch":"refs/heads/master"},"a412a1e447a87060c20aa1fa697c6ccf99d90152":{"kind":"REWORK","_number":3,"created":"2025-04-22 15:52:55.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/3"}}},"commit":{"parents":[{"commit":"d52e0532f978507a742504d4ab1dec3fa8473b4e","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52e0532f978507a742504d4ab1dec3fa8473b4e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-22 15:52:15.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport Current NetApp Cinder driver is not supporting https transport using self-singed certificate. This patch will enable transport using self-signed certificate.","message":"[NetApp]:Cinder support for self-signed transport\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a412a1e447a87060c20aa1fa697c6ccf99d90152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a412a1e447a87060c20aa1fa697c6ccf99d90152"}]},"branch":"refs/heads/master"},"405d55a5648cdb0352717b742f329b6be7c36db5":{"kind":"REWORK","_number":4,"created":"2025-08-10 14:34:00.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/4"}}},"commit":{"parents":[{"commit":"d52e0532f978507a742504d4ab1dec3fa8473b4e","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d52e0532f978507a742504d4ab1dec3fa8473b4e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-10 14:32:31.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/405d55a5648cdb0352717b742f329b6be7c36db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/405d55a5648cdb0352717b742f329b6be7c36db5"}]},"branch":"refs/heads/master"},"21a968ddde2bc0c0c29258897179691f4ef34c9e":{"kind":"REWORK","_number":5,"created":"2025-08-13 09:21:34.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/5"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-13 09:21:10.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21a968ddde2bc0c0c29258897179691f4ef34c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21a968ddde2bc0c0c29258897179691f4ef34c9e"}]},"branch":"refs/heads/master"},"ac69c376e81d9ace46fda4afbefeecc7b70ecdad":{"kind":"REWORK","_number":6,"created":"2025-08-18 18:17:41.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/6"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-18 18:16:03.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nImplements: netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac69c376e81d9ace46fda4afbefeecc7b70ecdad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac69c376e81d9ace46fda4afbefeecc7b70ecdad"}]},"branch":"refs/heads/master"},"431614e406d59672a550f89e5efff6a730efe6c6":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-08-18 18:18:50.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/7"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-18 18:18:24.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nImplements: blueprint netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/431614e406d59672a550f89e5efff6a730efe6c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/431614e406d59672a550f89e5efff6a730efe6c6"}]},"branch":"refs/heads/master"},"73062549457e05b2f783f37da3f6ed3f05449b9a":{"kind":"REWORK","_number":8,"created":"2025-08-19 13:32:07.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/8"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-19 13:31:53.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nImplements: blueprint netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73062549457e05b2f783f37da3f6ed3f05449b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73062549457e05b2f783f37da3f6ed3f05449b9a"}]},"branch":"refs/heads/master"},"a35be72ad84dd48d6d4cd2f5c4bb2d11e81cb5b4":{"kind":"REWORK","_number":9,"created":"2025-08-25 07:54:23.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/9"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-25 07:53:32.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nCurrent NetApp Cinder driver is not supporting\nhttps transport using self-singed certificate.\nThis patch will enable transport using\nself-signed certificate.\n\nImplements: blueprint netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a35be72ad84dd48d6d4cd2f5c4bb2d11e81cb5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a35be72ad84dd48d6d4cd2f5c4bb2d11e81cb5b4"}]},"branch":"refs/heads/master"},"f0db4b3282fb847cf4eff095b65a12022ebdd23e":{"kind":"REWORK","_number":10,"created":"2025-08-27 12:22:03.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/10"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-27 12:18:31.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nThe NetApp ONTAP driver now supports HTTPS transport\nfor management communication with certificates.\nONTAP systems use self-signed certificates by default\nfor HTTPS management access.\n\nThis patch enables two modes of HTTPS communication:\n- When ssl_cert_path is configured: certificate validation\n- When ssl_cert_path is not provided: Unverified SSL context\n  (encrypted transport but skips certificate validation)\n\nThis allows secure communication while maintaining ease of\nconfiguration with ONTAP\u0027s default self-signed certificates.\n\nImplements: blueprint netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0db4b3282fb847cf4eff095b65a12022ebdd23e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0db4b3282fb847cf4eff095b65a12022ebdd23e"}]},"branch":"refs/heads/master"},"d3d91d9a13f8266c3ac2489859ce14545792bd26":{"kind":"REWORK","_number":11,"created":"2025-08-28 19:10:38.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/63/947663/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/947663/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/947663/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/947663/11"}}},"commit":{"parents":[{"commit":"55d27f77a6e699738b8f96889d60b931a9fc4a9e","subject":"Merge \"Added support for NetApp ASA r2 (All-Flash SAN Array r2) disaggregated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55d27f77a6e699738b8f96889d60b931a9fc4a9e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-04-18 12:21:02.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-08-28 19:10:23.000000000","tz":-240},"subject":"[NetApp]:Cinder support for self-signed transport","message":"[NetApp]:Cinder support for self-signed transport\n\nThe NetApp ONTAP driver now supports HTTPS transport\nfor management communication with certificates.\nONTAP systems use self-signed certificates by default\nfor HTTPS management access.\n\nThis patch enables two modes of HTTPS communication:\n- When ssl_cert_path is configured: certificate validation\n- When ssl_cert_path is not provided: Unverified SSL context\n  (encrypted transport but skips certificate validation)\n\nThis allows secure communication while maintaining ease of\nconfiguration with ONTAP\u0027s default self-signed certificates.\n\nImplements: blueprint netapp-self-signed-https\nChange-Id: Ia486448b85feba636effef609d79ef8e57a6d39a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3d91d9a13f8266c3ac2489859ce14545792bd26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3d91d9a13f8266c3ac2489859ce14545792bd26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
