)]}'
{"id":"openstack%2Fcinder~947951","triplet_id":"openstack%2Fcinder~stable%2F2024.2~Ibc3f7b995c43bc281b6978bfcba4d1a50db41fa3","project":"openstack/cinder","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-10-21 10:38:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc3f7b995c43bc281b6978bfcba4d1a50db41fa3","subject":"Make volume Glance metadata bulk update idempotent","status":"MERGED","created":"2025-04-23 13:33:18.000000000","updated":"2025-10-21 10:39:51.000000000","submitted":"2025-10-21 10:38:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947951","cherry_pick_of_change":764045,"cherry_pick_of_patch_set":6,"meta_rev_id":"b14678bf49ce9d7842950b541e9f72d114a2df36","_number":947951,"virtual_id_number":947951,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-21 10:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-21 08:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-08-26 12:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-21 08:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-08-26 12:21:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 13:33:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2025-04-23 13:49:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-04-23 13:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-23 14:09:53.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-04-23 14:11:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-23 15:02:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-04-23 15:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 13:36:17.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2025-06-28 00:24:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-08-26 12:21:04.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-21 08:43:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5ff89bc91a132381317db3b9cc143fefce847325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-23 13:33:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf87e1cbf0cb4867a066a1f4c2e896a6c861792f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-23 13:49:18.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/762/.","accounts_in_message":[],"_revision_number":1},{"id":"5cfb9d659048b74e0f97f73abf1b2c441407ad2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-23 13:49:34.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/643/.","accounts_in_message":[],"_revision_number":1},{"id":"7e83bfbdbdfa243284632c8b5f7d7f3ceef77555","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 13:57:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e3fc3310464f5ba7449ea4b62039de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3697986bbc44482b81ae1f40a0c1fe82 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4715eca99e244801ab4ae5618afdf422 : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c656252b89b59e3f359b3cd9ffa786ac1190341a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-23 14:09:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/947951 : FAILURE in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ccd240fd66be78623b61f5007219cdb01e61620d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-23 14:11:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/947951 : FAILURE in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"607ed370940ec60223d55ad2f42d32b73586e24c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-23 15:02:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1e4e10e463aa43eaa91d6cfe58577a8c\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e7fb009dfcc4f948546ffb6c9412edd : SUCCESS in 46m 31s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/489e2cb0a1d2417d9cef87e5557f6583 : SUCCESS in 43m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/426a5a10fe5e44668d8dc27dbe86d5c5 : FAILURE in 1h 28m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81275f63979942e7983e6f26d6f0d247 : SUCCESS in 41m 43s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"71486ad8c9069246395abaf178bbc9e6bfdae8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 15:36:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55e9167bd09742acb4a1bc3d30575c39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486cdc55c0984f50802c9d172294ffaa : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25532b16fa7a4d07b9ae1b93eafef920 : SUCCESS in 12m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81a65b085eaf4b6f89a0afe4b1e4bfcb : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6b1d4e8292b4122ba270634cda04444 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9b3b98e9e94437afcedafbfb64679d : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/231442923e274269abb303549c9cc9c4 : SUCCESS in 45m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88ab2e42c1e34b5ca3d94f6cbe609ce3 : SUCCESS in 1h 27m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79bad62feab9452691567387d4f89b83 : SUCCESS in 51m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db96f5f087974cd99d082b12beea3ff7 : SUCCESS in 20m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b4329012afa2437a8da41557d816a196 : FAILURE in 14m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3fa3571eaa634a34ae33b2d70dd8d017 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f739912d88a4dc29baa4eae28b779c8 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/77280193b85b48fdbeaae61867d1242c : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2e014b337c964ad0a132ef84996d2805 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc6b383b474148b8a145f7a7002be7ef : FAILURE in 33m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50bd30b8ef214726976abcb734296683 : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55b42fe1e52a4bdd9b7dc2deea9d5734 : FAILURE in 1h 46m 41s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08023794c7bf4de995f626ca64494e2e : SUCCESS in 1h 49m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a2636459aae24b3d8c9784e12f6b661d : SUCCESS in 1h 55m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/696a976708704f2c850dd25fe968be3c : SUCCESS in 26m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8fad02aa41c40f08375c40b19bbe9fc : SUCCESS in 1h 05m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cefd68c7725d448892aae4fe0278b7db : SUCCESS in 1h 27m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/427d87558a3c4244a766c70cacd86aba : SUCCESS in 38m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c5991bfd1b9437887bf5806880096e8 : SUCCESS in 1h 26m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ccb9e6d1371042f9bd595ab54c2ef44f : SUCCESS in 1h 31m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67d1ce94dff444bf94d79393b7d79fd1 : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"13fad32b582754b37bd146dcc44f19ff7ab27711","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-06-16 13:36:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcf54b14db5908cd6ebe291de9fbb246e2a3db5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 13:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/462c43f32ed24ea28dfb5e5614815a73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a0c901a88634dc8baa0e1dbac315b41 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d804ad4b8b384258b0773909c0061e10 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec70442ca51169f82246ab5d3257d3176c92c71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 15:40:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24cdf2e7556849fcb3dac60239d10dd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a322797b6bd54cbaa658ba5bd7727f67 : FAILURE in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8847b4c5cb3741d49f29a35942d21edb : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43e6470759d04cf69faddda6ba33a005 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b2f3d971c244c98be5e87a6549d3c85 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30dec3442eaa437fa6e6ee94820e9080 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/2617fc66bd03437b844e971d870b9502 : SUCCESS in 49m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/949fcd98803d4ce397eff1fec4fb98bf : SUCCESS in 1h 27m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3da2ca5969249e6ac0db0737bec3119 : SUCCESS in 56m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/518b288347d847498f44b8699cf00d71 : SUCCESS in 21m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8b8752ddc2a4eb7b789e25389a6f0d2 : SUCCESS in 17m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3526df5a3a1f4b788da4e895100c24a2 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ff4cc92d45a496db9eaa96cd8c22fba : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43a4773764454759968c6189bcd407d3 : SUCCESS in 6m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bbbb17ceaf7d4339a83ee25e70757e49 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cd50635daa22482aa6fdfe2c43946fd2 : FAILURE in 23m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/452ee3f7017e463e8ae9271a19038344 : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ecf57abcae2048aabedc3649f987aef0 : SUCCESS in 1h 39m 03s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c208ff99706b4fd6a7a334cba3f31b2f : SUCCESS in 1h 59m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5bbfdce9c10b40239ed9c932fc3edaea : SUCCESS in 1h 46m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a91a7b916b834928946133804eb27b88 : SUCCESS in 28m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76bbe5b3326d4740824715c0ea2ab236 : SUCCESS in 1h 07m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1483c73df2bc4998a573e84592ff15bb : SUCCESS in 1h 20m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ecd264402a024372936286cf3ea3420b : SUCCESS in 37m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f258b87ff0b4e619d66a174f6405fc4 : SUCCESS in 1h 18m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e3965d1984624c9dafd07cb9888b7421 : FAILURE in 15m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bc0c2b82a224d70ad577187f77808f5 : FAILURE in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9cbfff6286680f7a49a80b7d3c3b84157642d0d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-27 19:11:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1a8b236d698196a51cfa4f79fdcf2442ea001c5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-27 19:54:26.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/96a357b3cdeb442b94aca2713704fdae\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd2e7c966c9b47fb9a939c306ffb60a8 : FAILURE in 22m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d8ff2d93cd845378fe58515c37defbc : FAILURE in 22m 49s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/670fd30f2a9a4183adc2bbdfb8d9f7f0 : FAILURE in 21m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1d59f4352281f10a93e87376ede3a68ca86f7495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:15:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992399dadf9e4089b49d17074580ddda\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9acbcd1a3d047ab899fe12a08f278d3 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/741280349faa40a599261c767f6bebd3 : SUCCESS in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ad4b812405a9a8032de5384dfa992bc3c2ede3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 21:22:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d69ff30191a544bf8591e9be48b1fd55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c191bd5f3ca4f359cde4b2843f97f04 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58436543c28246d9b80051b315dd952e : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e31a33742214e53bcfe4ce77f07785d : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f1ca97cd73430abc0a6f07c0ab5b1d : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08874e3da8a14c79aa4a7b6a4b0ff5db : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ae9b2ce6b1e849c1af56f1e4bf93d7b0 : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/025c2554d0e342d290d82d42f28faa1f : SUCCESS in 1h 25m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90155420fad045c49765fe09f8264817 : SUCCESS in 31m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c51925ca867944ae864220e39c3cb815 : SUCCESS in 21m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b48f27879534ce380d50b1a163077b9 : FAILURE in 14m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/82a274df72074118994df68140fc3a88 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4543d0e1e8e0481cb48a425d639c2c89 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2461273efcd4d5f80e5cf79bf1ecbb0 : SUCCESS in 3m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8137fba9b8284e678b458ceb88fa48cf : SUCCESS in 4m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/49a7c610d4f44df786613cab56342982 : FAILURE in 31m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23289132fc054621b8d3bbb7f56206a2 : SUCCESS in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64f8527778e7468da14f4e9d65b10e12 : SUCCESS in 1h 44m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45c0fe3bcac9473699e5a9e531e398d8 : SUCCESS in 1h 58m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/aca059c1c65c4ce2a9e4178fd776b58b : SUCCESS in 2h 10m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7612265bb311474db11c1dcf2e184f79 : SUCCESS in 14m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a0729ad5d6a7418490a6b875c78de6ef : SUCCESS in 1h 08m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/95aad36b88bf465cb39e74ea184f52cf : SUCCESS in 1h 30m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac09a4245de34df8aed91e11844e5edc : SUCCESS in 37m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48a1c8985414407da7ee3c8f21253e7c : SUCCESS in 1h 28m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a0e71014d48242c9a73bd6ade9da86b6 : SUCCESS in 54m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65f3daa47ecc4b11961c8cb358c7d9fa : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1a5072cb5e537c84a5ef16427e6424ab50180f00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-27 21:46:56.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1098/.","accounts_in_message":[],"_revision_number":2},{"id":"567d96e330f6c2f6447da8f9718303defb756899","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-27 21:51:21.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1227/.","accounts_in_message":[],"_revision_number":2},{"id":"a22fe44b41a060b1a9c32594f50bd75642175eda","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-06-28 00:24:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f108b74476f613ad30f81fec6d57d1f761910fdd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 12:21:04.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c43f687bb6ac23421a90261ac3f64e7607ad593","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-21 08:43:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae856bc32aabbb1e075288a104a54415023cab0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 08:44:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15c735152fd6bdd2f9f052a7fef9214d9e66f07c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 10:38:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b6cea461c044908216dc742aa0499a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186ab4f615a348bb9f18fbb33f4db029 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31a732694bcc4e9fbeab0c196ba714aa : SUCCESS in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/510f2dc9aeab4b45b8ed495c958e14ca : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e904b7ed81d4cfabe3b294596220ded : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5c62fadfb35f48e499154e1457c6bcd5 : SUCCESS in 39m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0f3c80b31cf42238324297c71c41f44 : SUCCESS in 1h 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa50a436c9364787a0354b1438798442 : SUCCESS in 55m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53632f781f8c49049845bbcf32ff5fa8 : SUCCESS in 15m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60971b511edd47f8a4abecbc4f3db0cf : SUCCESS in 1h 07m 27s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5a22db3a9c941489d1903124fec1856 : SUCCESS in 1h 52m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97ccc52a5cd241fb9fc7ca6457a7820f : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"26dfcc71922c971acec68ede7603e438b606571f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 10:38:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b14678bf49ce9d7842950b541e9f72d114a2df36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 10:39:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595befe07b064e3283637ff63a20173f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48de138e8dfc4d199b772dd107a76088 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a1562d036b48d695ce00dcc0c94d2f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97eef845d4101e1c142d77dc37a10d0e04e7e336","revisions":{"d1464633aa93ba334f4187744d3c65c78ad7902d":{"kind":"REWORK","_number":1,"created":"2025-04-23 13:33:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/947951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/947951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/947951/1"}}},"commit":{"parents":[{"commit":"c579760ddaee4e65dc45f8f546814763af0da0ef","subject":"RBD: Fix upload volume with different format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c579760ddaee4e65dc45f8f546814763af0da0ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 17:36:19.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-23 13:33:18.000000000","tz":0},"subject":"Make volume Glance metadata bulk update idempotent","message":"Make volume Glance metadata bulk update idempotent\n\nThe cinder.db.sqlalchemy.api.volume_glance_metadata_bulk_create function\nhas the following in its docstring:\n\n    This API does not support changing the value of a key once it has\n    been created.\n\nHowever, it actually fails if a specified key exists and has the same\nvalue. This can cause problems if volume creation is retried following a\nfailure, since the metadata may have already been created.\n\nThis change modifies the function to ignore keys that exist and have the\nexpected value, as per the docstring.\n\nRelated-Bug: #1823445\n\nChange-Id: Ibc3f7b995c43bc281b6978bfcba4d1a50db41fa3\n(cherry picked from commit fbf4899764b4ada68e42b39cec80a94c95cef03a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1464633aa93ba334f4187744d3c65c78ad7902d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1464633aa93ba334f4187744d3c65c78ad7902d"}]},"branch":"refs/heads/stable/2024.2"},"97eef845d4101e1c142d77dc37a10d0e04e7e336":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-27 19:11:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/947951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/947951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/947951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/947951/2"}}},"commit":{"parents":[{"commit":"73f25c6886e87a37d628867ed025fc1b12635cd5","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73f25c6886e87a37d628867ed025fc1b12635cd5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 17:36:19.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-06-27 19:11:05.000000000","tz":0},"subject":"Make volume Glance metadata bulk update idempotent","message":"Make volume Glance metadata bulk update idempotent\n\nThe cinder.db.sqlalchemy.api.volume_glance_metadata_bulk_create function\nhas the following in its docstring:\n\n    This API does not support changing the value of a key once it has\n    been created.\n\nHowever, it actually fails if a specified key exists and has the same\nvalue. This can cause problems if volume creation is retried following a\nfailure, since the metadata may have already been created.\n\nThis change modifies the function to ignore keys that exist and have the\nexpected value, as per the docstring.\n\nRelated-Bug: #1823445\n\nChange-Id: Ibc3f7b995c43bc281b6978bfcba4d1a50db41fa3\n(cherry picked from commit fbf4899764b4ada68e42b39cec80a94c95cef03a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97eef845d4101e1c142d77dc37a10d0e04e7e336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97eef845d4101e1c142d77dc37a10d0e04e7e336"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
